)]}'
{"id":"openstack%2Fnova~1002658","triplet_id":"openstack%2Fnova~master~I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a","project":"openstack/nova","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-31 22:35:38.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-31 23:09:28.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a","subject":"neutron: Simplify client creation","status":"NEW","created":"2026-08-27 14:49:37.000000000","updated":"2026-09-01 01:35:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee74cb89d4c16ecc95bb4893d387a03228de12ee","_number":1002658,"virtual_id_number":1002658,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-01 01:22:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-08-31 23:09:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-27 14:56:56.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-27 15:44:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-27 17:47:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 01:24:53.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"9d9d9826dfc710b547e9bb151d1dc049da7459ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 14:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abfc030ccfc8d7b1fdeb8baaaea2af439d435bbe","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 14:56:56.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ecb2481f8efa4709af24957a9173b9ea\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a3fb13a91c9e48959e596d938b99e5d1 : FAILURE in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1c051539a7be9e7109a22a00287c8f4bfa2ba3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 15:23:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"295856af9bb1e1e8aac0f64c770d6b0b57fa1d9f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 15:29:05.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3131e8d75ed845ada91c472fcc8c0813\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/13f6128aac034ad1b7d0f038efad1930 : FAILURE in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0dcd61daba4e3d2b6f7689621d8f41d874362458","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-27 15:44:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a .","accounts_in_message":[],"_revision_number":1},{"id":"9841a5994e393fd2732276ad0682cd9b39787931","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-27 16:19:53.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a .","accounts_in_message":[],"_revision_number":2},{"id":"7a00342dd8e768b352980c8e69e22e623c212c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 17:47:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1f3218df160443ca26cf2f533f1d05b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4804aed91357474daba5da73aee3a621 : SUCCESS in 46m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/314f1e360e6d48d59a62c80a298d1a25 : SUCCESS in 1h 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/852389e57fed4df288fa81074228aad4 : SUCCESS in 1h 04m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c13cf93406054c8ca35c8c60dd950b1b : FAILURE in 49m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de961712cb54545b8ebbd7a7c310e78 : SUCCESS in 9m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f55b834a8ca4a18bce7d13624231c8e : FAILURE in 9m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a37bb5589d504547a773a42badcc1b2f : FAILURE in 8m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc9c4f17248a47408a8e4c004f40664d : FAILURE in 17m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad445efce77c4a9a888be4047934065c : SUCCESS in 15m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418a8dbe1064459a9a67e87aebc123d4 : SUCCESS in 9m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/79253e341aa3433a9acbbcdf9e8f0792 : SUCCESS in 30m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ccefcac4b0d04cb8a2156b6fdafc4116 : SUCCESS in 1h 47m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7234ed6aecd144068d4df0ca73f81b6a : SUCCESS in 56m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/7540e9ec20b74306a2d103808a6a6712 : SUCCESS in 37m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/cf571a3c1fc74462b46c5d7357fee758 : SUCCESS in 1h 00m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e93742825cf14c158f35f3347bcb5f8e : SUCCESS in 53m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/929c9ab26a324cd79007caf265537de8 : SUCCESS in 1h 43m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/b9366d79e66f4d42aacac59ab066f180 : SUCCESS in 2h 17m 10s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5e6321d2aa6c4691bf1bc8c6cb209d97 : SUCCESS in 51m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dca837fe6d9f477b9398b27e3c920eea : SUCCESS in 6m 18s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bcebda8874c94f8a8e0bf96ad8fc0792 : SUCCESS in 26m 09s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/445eb1d28e7d411ab977c9b9def7b0a6 : SUCCESS in 29m 25s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/263b80597d5745328fb5fd15713e2b56 : FAILURE in 35m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1790e8695972414da0875c787df53057 : SUCCESS in 1h 23m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4fdfcf9589848ef827940caa489696e : SUCCESS in 56m 07s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/820944b67b6e4cd2b0e2d32ae60b1963 : SUCCESS in 1h 00m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/09c8f9de1649455199af7b588cf58254 : SUCCESS in 37m 06s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6bf5d79c08a545e4817f8762c8a86a1d : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7bc28ccb8aff20df9a81862a0b5fcc286f7e1632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 18:55:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4bc54888d84400b447a1699dd0fa96\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/368b2829c0d644438c5e9611f2252161 : FAILURE in 15m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9124ad065ce84ddfaa97fae121d6ff44 : FAILURE in 12m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1dc4adf67174b7cbf7f346ef104cf39 : NODE_FAILURE Node(set) request 562f0061ba154a7b800b024e0bb06754 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dde89e673bd1099e9ff44af8e45888655f82f105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-31 22:29:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"84d6157f7da0a0e8c36df66fcf594b9a00d23a77","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-31 22:35:38.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ee51bb19f31e43c0b2b3d0b6fd7a9c9e\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/607265be7c0c43c0bc6d63e250c55fd1 : FAILURE in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa94e80f5386e039f82b758e51a871408d410e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-31 23:03:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d8fe91c455a7611828e152200c5f049c97ed2eb7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-31 23:09:28.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/0ea658f8035446a4ac346821755cdf7c\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/009954319e18423385fe87e79a92c7a5 : FAILURE in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ffb2bf98b9904ac685544d33da89b637f7c1787d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-31 23:27:34.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a .","accounts_in_message":[],"_revision_number":3},{"id":"c09f08a6bb6d1b58f66ae2fb4d21071c1cbf252c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 00:24:12.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a .","accounts_in_message":[],"_revision_number":4},{"id":"7b7524a72312d856a986af9c4fdc01279fda9229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 00:33:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68703be403c749cfb248f6e588f1488d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d36c6425d84646a9974632c5caad87b6 : SUCCESS in 27m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad18f22aa97e4258932cc99fa368a357 : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/62abd5a05249437e9ea6ce19324773ef : SUCCESS in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da9dbde60d8553a6294a9ed5e10d2558094dc293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 01:22:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a42b874f63416fb6ac0f12772287f0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6801e35a8fdf449bada06e7594e9b283 : SUCCESS in 1h 01m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cf839210aba048e2ae19d84d0dd9be79 : SUCCESS in 1h 39m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7604c80eb6e3490da4a4c931eb09684a : SUCCESS in 1h 03m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d13b12126747c098bcb9405afedd91 : SUCCESS in 1h 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c309171cc78489ba241ace9045870b9 : SUCCESS in 8m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fbda7ba6a2545f99a4a2296181748e0 : SUCCESS in 13m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f99328f851f049abafa3bec63031b528 : SUCCESS in 14m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a0cda529186949f0b3951603df3acc52 : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa50e76641845c280a3a7513cd71670 : SUCCESS in 19m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5213024952d942ba93c0f68e9391c88d : SUCCESS in 5m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f912d31d806b4e8486fa013b52549114 : SUCCESS in 1h 06m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6f60075076bb4381b24105213ac9ad68 : SUCCESS in 1h 59m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/241833a69a8c4287aa55d7fbd58b1163 : SUCCESS in 48m 29s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6ad041edbe284bedb865cbebc854426b : SUCCESS in 37m 22s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/da7cdd45dba2476abe4a6ba1306c3845 : SUCCESS in 43m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/abd956d8a11a4373aa664e09e1455687 : SUCCESS in 35m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/69912e76ff144d899c83a728d5b7db0e : SUCCESS in 55m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/1eee93bb519f49909bb683a247211c18 : SUCCESS in 2h 10m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6fe203ca879c46d388915763f4ffb2af : SUCCESS in 53m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1921320e30db47e7a26b077142a881a1 : SUCCESS in 6m 48s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/831ffac9a2754a82bfc3b6680271308a : SUCCESS in 27m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/09eb192b161c4952a8d27df1112be4cd : SUCCESS in 27m 39s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/176754e1a6c7447c94ba7998f97c9bc6 : SUCCESS in 36m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/48ef2bd886494b36b4feef3385736f63 : SUCCESS in 1h 29m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9795106f5ec44fe4b729a5d37ded8275 : SUCCESS in 1h 13m 28s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/901c93f1c02d4a43827cc827dfbdd504 : SUCCESS in 53m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/522d8147b7844ef8ae0a3923a377af51 : SUCCESS in 33m 53s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d8a7f7a9621f4109a5005fca9497d921 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":4},{"id":"ee74cb89d4c16ecc95bb4893d387a03228de12ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-01 01:35:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b8df6c75fa9c36b68019297f6051503185102566","revisions":{"6d0ca7f09fb35d285e8de6e0cafb197e5ae76471":{"kind":"REWORK","_number":1,"created":"2026-08-27 14:49:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/1002658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1002658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1002658/1"}}},"commit":{"parents":[{"commit":"69fe11a0af9cd1169cea3bf883fbb2ffb3c987cf","subject":"conf: Deprecated http_retries options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69fe11a0af9cd1169cea3bf883fbb2ffb3c987cf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-27 10:12:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-27 14:48:57.000000000","tz":60},"subject":"WIP: network: Simplify client creation","message":"WIP: network: Simplify client creation\n\nPer the note from efried, we were building an adapter in order to pull\nconfig options, which we then passed to neutron which built a new\nadapter. Turns out this is wholly unnecessary: ksa provides the\nutilities we need to pull the oslo.config options into a dict for us,\nwhich we can then pass straight through to neutronclient.\n\nAs a bonus, this should make the ongoing migration to SDK easier.\n\nWIP: while I investigate the test failures\n\nChange-Id: I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d0ca7f09fb35d285e8de6e0cafb197e5ae76471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d0ca7f09fb35d285e8de6e0cafb197e5ae76471"}]},"branch":"refs/heads/master"},"cb581ca2d16c3bc976b1c05a039768fa431bcb7d":{"kind":"NO_CHANGE","_number":2,"created":"2026-08-27 15:23:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/1002658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1002658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1002658/2"}}},"commit":{"parents":[{"commit":"1939a399c045178b6b334ca61115ee040ecb22fb","subject":"conf: Deprecate http_retries options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1939a399c045178b6b334ca61115ee040ecb22fb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-27 10:12:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-27 15:23:06.000000000","tz":0},"subject":"WIP: network: Simplify client creation","message":"WIP: network: Simplify client creation\n\nPer the note from efried, we were building an adapter in order to pull\nconfig options, which we then passed to neutron which built a new\nadapter. Turns out this is wholly unnecessary: ksa provides the\nutilities we need to pull the oslo.config options into a dict for us,\nwhich we can then pass straight through to neutronclient.\n\nAs a bonus, this should make the ongoing migration to SDK easier.\n\nWIP: while I investigate the test failures\n\nChange-Id: I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb581ca2d16c3bc976b1c05a039768fa431bcb7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb581ca2d16c3bc976b1c05a039768fa431bcb7d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"69fe11a0af9cd1169cea3bf883fbb2ffb3c987cf","ours":"6d0ca7f09fb35d285e8de6e0cafb197e5ae76471","theirs":"1939a399c045178b6b334ca61115ee040ecb22fb","merge_strategy":"recursive","contains_conflicts":false}},"e34fd3e2145631dfb1f63c6de28a784a486041c4":{"kind":"REWORK","_number":3,"created":"2026-08-31 22:29:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/1002658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1002658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1002658/3"}}},"commit":{"parents":[{"commit":"1cd6d6a3458c905d8300b82978d2d4cd6ddf479a","subject":"cinder: Replace legacy auth options (3/3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cd6d6a3458c905d8300b82978d2d4cd6ddf479a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-27 10:12:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-31 22:16:44.000000000","tz":60},"subject":"neutron: Simplify client creation","message":"neutron: Simplify client creation\n\nPer the note from efried, we were building an adapter in order to pull\nconfig options, which we then passed to neutron which built a new\nadapter. Turns out this is wholly unnecessary: ksa provides the\nutilities we need to pull the oslo.config options into a dict for us,\nwhich we can then pass straight through to neutronclient.\n\nThis isn\u0027t an identical switch: we get a couple of new options passed\nthrough here. Previously we were explicitly passing the following:\n\n- service_type\n- service_name\n- interface\n- region_name\n- endpoint_override\n- connect_retries (explicitly, rather than from the adapter)\n\nNow we pass the other options set by process_conf_options:\n\n- version\n- min_version (a no-op since we stub this out with a fake opt)\n- max_version (as above)\n- connect_retry_delay\n- status_code_retries\n- status_code_retry_delay\n- retriable_status_codes\n\nAll of these are desirable though, since we were already registering\nthem: we just weren\u0027t consuming them.\n\nAs a bonus, this should make the ongoing migration to SDK easier.\n\nChange-Id: I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e34fd3e2145631dfb1f63c6de28a784a486041c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e34fd3e2145631dfb1f63c6de28a784a486041c4"}]},"branch":"refs/heads/master"},"b8df6c75fa9c36b68019297f6051503185102566":{"kind":"REWORK","_number":4,"created":"2026-08-31 23:03:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/1002658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1002658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1002658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1002658/4"}}},"commit":{"parents":[{"commit":"34c107a165349d2c9c59e84557d90606a95fe632","subject":"cinder: Replace legacy auth options (3/3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/34c107a165349d2c9c59e84557d90606a95fe632"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-27 10:12:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-31 23:03:07.000000000","tz":60},"subject":"neutron: Simplify client creation","message":"neutron: Simplify client creation\n\nPer the note from efried, we were building an adapter in order to pull\nconfig options, which we then passed to neutron which built a new\nadapter. Turns out this is wholly unnecessary: ksa provides the\nutilities we need to pull the oslo.config options into a dict for us,\nwhich we can then pass straight through to neutronclient.\n\nThis isn\u0027t an identical switch: we get a couple of new options passed\nthrough here. Previously we were explicitly passing the following:\n\n- service_type\n- service_name\n- interface\n- region_name\n- endpoint_override\n- connect_retries (explicitly, rather than from the adapter)\n\nNow we pass the other options set by process_conf_options:\n\n- version\n- min_version (a no-op since we stub this out with a fake opt)\n- max_version (as above)\n- connect_retry_delay\n- status_code_retries\n- status_code_retry_delay\n- retriable_status_codes\n\nAll of these are desirable though, since we were already registering\nthem: we just weren\u0027t consuming them.\n\nAs a bonus, this should make the ongoing migration to SDK easier.\n\nChange-Id: I811cfdeeb00bf01bbe1a55de97c068f08ee97e5a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8df6c75fa9c36b68019297f6051503185102566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8df6c75fa9c36b68019297f6051503185102566"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
