)]}'
{"id":"openstack%2Fnova~1003266","triplet_id":"openstack%2Fnova~master~I6f34ad32a4c5eb80131673be1572fe2fd609bcad","project":"openstack/nova","branch":"master","topic":"bug/2166009","attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-02 17:50:23.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-05 03:22:16.000000000","reason":"A robot voted negatively on a label"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-04 06:43:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6f34ad32a4c5eb80131673be1572fe2fd609bcad","subject":"[func test]Catch hanging task at graceful shutdown","status":"NEW","created":"2026-09-01 12:30:17.000000000","updated":"2026-09-05 03:22:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"1c0ce8d7ffc5fcf3bc5610afb037e68ec5adf2e9","_number":1003266,"virtual_id_number":1003266,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-05 03:22:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 12:37:21.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-01 13:24:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 13:40:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 14:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 05:11:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"03b21b59acc18efa90e2e2d894f5575c9d495602","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-01 12:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e95e1acddbdf5a271b9f6c3e0bba0565bebe469","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 12:37:21.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/0759d6b261494af88dde9d08ba98c26e\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/958b48f14ce94499b69790740bfeac90 : FAILURE in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72032d8c92f591c3f4a702be59fc7899c31110b5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 13:24:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":1},{"id":"1109c6b71071ee6a1defcffc0eafedc7082fefcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:40:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b83fc788ba4d4871a5c2e229c505c4cc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9967319902be4ca9ae490e1429c9221a : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22ca41725cb049858ac2068415b63ea1 : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0a4d4e635284e08aa97436c96a4553c : FAILURE in 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccd4b067ed2df947c9dc033839ab9940bfbc5627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:18:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/627be88a17fc4168bf2577af73af7f10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b15d27e19d84e18adf96cfa9b55c32b : FAILURE in 1h 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b7a8a7f8d041a6a18c520bb70bbfb8 : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c21d5c9afab406394b839201376ad69 : SUCCESS in 8m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b16938ca7b3a49f98ab27e6c59422b3f : SUCCESS in 14m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97fab24962e3464eab6548a08e6fde6e : SUCCESS in 17m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da4e600c79a45439644448c6ed24b01 : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4308049f4924d9fb09ca323fda3f604 : SUCCESS in 3m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/81125066db5a405bafdae13179aa8208 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5402b58433494cc19631ebdaf76f9fd3 : FAILURE in 29m 01s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/130075c8bc3c42beab875aed07269cf2 : FAILURE in 34m 47s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5b4f2ec87f734fa484c346132897236b : SUCCESS in 43m 33s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a2e7f90f5e0a480b9280d8476f65bf87 : SUCCESS in 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"26199ed8c6e4d884248ce20b124aa44422bc8291","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-01 15:04:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dfec2cfd7cad49c79ac7906208a152aa78b2a3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-02 17:50:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9eab5b6f56b0edc25b5ae98ef64b9bcbf3bb73cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 18:02:56.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":2},{"id":"0d47fc1fdf18ab59eb5f07608c5c5b3d9e606f0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-02 18:49:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c8636a54be22670a0dfdf63dab85d7645c07338","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 19:01:24.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":3},{"id":"5e151701a0f5eb741d34d22f7822576f788cc908","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:25:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac5264cfd644fab9ae6e27090245d48\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0110287671a74cb9bae74f8783e9def9 : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56f9266c55ee42b38da8ed0f07d8ce9b : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/27d782e400bb4b3bb8198beb0b6594f1 : SUCCESS in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b021adde29497430ee6898f759bdd8bb984c7589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:45:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06e1ca126eb54555b74a3d69ce060ba0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cd6bd9dd0f542b786669d61695e7fc3 : FAILURE in 53m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dac25e565414fe7ba34ca9c3ff7d2d0 : FAILURE in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74e2dadf7485439aa49ad4c77b542cb7 : SUCCESS in 11m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c12fcde0152b4e7c9c2b4acf022353ee : SUCCESS in 16m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe20ca47887d417984c47034c124af1a : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff7035b91c6043e78c895a5dc559e372 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60f5bba7b68d40d89ab7aad25c844fb5 : SUCCESS in 4m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/51a9b7548f774e48b9a2ece23e9aa1c1 : SUCCESS in 3m 49s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fe2148e3fc464670ae9ef9df5d02dd3c : FAILURE in 33m 13s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/07c9e48809c84514a48a6fae4fe30dd3 : FAILURE in 34m 58s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/caeb5d6116914d53a85b3627f5b02001 : SUCCESS in 37m 03s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/83f4f37114864356a437a55fe521441e : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":3},{"id":"d2833ddb67713b5432a4e7a2313be87a8ca01711","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-03 08:54:19.000000000","message":"Patch Set 3: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"29685a4e80fea73fbc123b1ccde60b2f323d6b0f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 05:11:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3db34361edf8dde5069f54b6e869f9a32b46ddb5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 05:12:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fda5cb6ef5d26555df78d9684b60fc246c39f2b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 05:13:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8881bc41c4534030a8cdb5bbb7609dd2ecc84604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 05:34:18.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"681808fbc22ae7552b1ffcdf741320956fdc948c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 05:59:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf59e416bcc3db4d55c087aae29b9dde4dd94fa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 06:13:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94ab06eb795c4e429e2439ea0e21b4f0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ed79b9c21374599a9f5abbc179a8b6c : SUCCESS in 22m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d5ce29d501e46c592e215ab5c0420dd : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cdaaae1840724e2785bbcf494d033982 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a5b3f125a189df8a363e6554f14089fdb1087101","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 06:30:58.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":4},{"id":"3d89ae0b90b8bd8428c3fcd34aa36da1537bd716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 06:43:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49cb39e8eb89431ab7e6ac67c96c8496\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e23766a4ca4540d68a1154c72e9b2ae5 : FAILURE in 1h 08m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7323b8429d764acb9e28b39c109a76f0 : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2d3c6f3a3394c7995309077bd10295d : SUCCESS in 8m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdba8336bc9d4684a6b7748ed272ee17 : SUCCESS in 8m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/630f9f7d763346f2b00ee7a2bed23e11 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db17ca2f2d9b4df198c37b8075221e13 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb88d89f243148b18525cca3b843a237 : SUCCESS in 9m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7340d254111142a4a616288a96e80f39 : SUCCESS in 3m 20s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d47a2a12cc5e4315974ec5f94dc0a7a9 : FAILURE in 25m 33s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c50cb274e3284abf86186b92c272a960 : FAILURE in 35m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/20468d59813d42de9158915eb4cdd125 : SUCCESS in 35m 46s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/48536e42895143ba80a715eedf3fcb51 : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":4},{"id":"bbb7562f3d40138b66cd9e4c5bdcc66699ff9fa6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 15:27:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df5e57c424cfbc7f224150cb34e416b3bbba30f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 15:29:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f9c06b0934f6557cac8405282849d33e7b61c88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 15:30:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d35e47d4f6427009b6c721a3df4ec534cb2397b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:51:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93b4e844725748c18ef4278e0c0e405e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d4e1ca499d7643d4a84458d2c6d3ad89 : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b01c6fd654a4d30a54f1f7235dab6c2 : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/49f11c6d515944b2a0023b05029e4783 : SUCCESS in 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"344779095adb599632b3e03af4203152a34ba75e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 16:27:03.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":5},{"id":"b0f3dbf35a46db1b19d613efafd6b0cb7406890d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 16:51:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"2a42f2a487bb7b4e3fb2c526228a01a9550be478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:16:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97666d6625a4a7c942ed95d5164cd57\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b841c962cc1b4887be7ae9cc02657b37 : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a59ad34603514158abd6ac737de840c5 : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d44dd581705043a4a40e07f4ded6b1f6 : SUCCESS in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e64d129598dba297c8f3e1151d509255333c1f32","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 17:47:41.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":6},{"id":"e65ab707afd3f4006715bf23f9508277a5002099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:54:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31ced02aaf5a4e9cac20ca4203915792\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c8365cb9dd4ff0bd88a3f1438ecac9 : SUCCESS in 1h 24m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e86c2a78c5410d8c8476d29963d60f : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3ab1d5e475e4ddab1a45faa9b2e58df : SUCCESS in 14m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/742554356f0a458cb9a46755c01395f5 : SUCCESS in 12m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d63564d524b8459b8867e719d37512bc : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31ea8f1a0364d8699ee93d79ba3b23b : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f1f5446cded478ea8d87b4e8c6bd885 : SUCCESS in 6m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8b7fe8256c924928ba8b035869a5863f : SUCCESS in 3m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4ac3dadeaa6c4c1d9e5fedd8c94f3c2d : SUCCESS in 33m 12s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9410adeb77114081bc87724694c1f476 : SUCCESS in 24m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/339fa36ff15e40fe82741bafa06d48d2 : TIMED_OUT in 2h 01m 13s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d481caf10912404eabf6f608301d5764 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":6},{"id":"5f690574566ddd72a30dd1b726d76835e98a40b4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 20:34:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"51144d069500fd87922f515fd3d9625379013eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 01:16:19.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"9c48712a73f01453b932d39f2a5350f423bb70d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 01:38:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77b53d226d8f4baea892897d989b4925\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c1e63bbd08e42bcb47ee1adc5e4e736 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/66c2493c0a1a4347bd7fd595c316f91e : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/192fa8ae52c04f6e89ecf3ed2b9d931e : SUCCESS in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dbbebce44d5ed24596fab4056922ece0025e6214","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-05 03:01:45.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6f34ad32a4c5eb80131673be1572fe2fd609bcad .","accounts_in_message":[],"_revision_number":7},{"id":"1c0ce8d7ffc5fcf3bc5610afb037e68ec5adf2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 03:22:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c05527ec6eb4eb49f777db70a58f762\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b11b7ed07474497fb159c6e5c9707cdf : SUCCESS in 1h 00m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b79ebaf76354e0abd9a671af4507a47 : SUCCESS in 7m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e40844343b34e8299670cb4aa00fd56 : SUCCESS in 10m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddc874fee8bc45798f1f4544f58f4e0b : SUCCESS in 13m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8ba8fbe21f5d4ce690e7cc2ffee4423d : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c524179031b44fcadb120e5f96c1a55 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97953e65f9ed436d989b8fd291cc5d79 : SUCCESS in 3m 59s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b3723920e84c4e8fbc21c967495a95f8 : SUCCESS in 8m 06s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9018cb66df394561b18991721a526e2b : SUCCESS in 32m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c9a041c3e72b440f8a3818f6cc565de1 : SUCCESS in 25m 07s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/de2c0cbbc78b4d5696557c2bef9cc043 : TIMED_OUT in 2h 01m 17s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ee14298271794e3bbbaa203b88401d11 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"05d9a08244c15f30ff3257a8ca40ffd07ad35d80","revisions":{"2cec50f11fe9efffdc013ed5a4a017f92e7df31b":{"kind":"REWORK","_number":1,"created":"2026-09-01 12:30:17.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/66/1003266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/1"}}},"commit":{"parents":[{"commit":"97e24fed386a4eb3945e1129937685241be2461c","subject":"[test]Ensure shutdown thread is waited on","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/97e24fed386a4eb3945e1129937685241be2461c"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nThis shows 3 tests with hanging tasks each of them is related to on\npotential task tracking bug in graceful shutdown during live migration\nrollback / abort.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2cec50f11fe9efffdc013ed5a4a017f92e7df31b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2cec50f11fe9efffdc013ed5a4a017f92e7df31b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"97e24fed386a4eb3945e1129937685241be2461c","is_merged_in_target_branch":false,"change_id":"Ieb5843980613c492de9da2a1cbefff45d8c17687","change_number":1003245,"patch_set_number":1,"change_status":"NEW"}],"branch":"refs/heads/master"},"16b5ff8a4c6f6b0ab21b85e5d359099368719d04":{"kind":"REWORK","_number":2,"created":"2026-09-02 17:50:23.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/66/1003266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/2"}}},"commit":{"parents":[{"commit":"9d2b3b14f33570ec28396be9c46d0267028a143d","subject":"Fix task leak for live migration rollback case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d2b3b14f33570ec28396be9c46d0267028a143d"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-02 17:49:54.000000000","tz":120},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16b5ff8a4c6f6b0ab21b85e5d359099368719d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16b5ff8a4c6f6b0ab21b85e5d359099368719d04"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d2b3b14f33570ec28396be9c46d0267028a143d","is_merged_in_target_branch":false,"change_id":"I363c051ce112a28558214964ce276e47fe521e6c","change_number":1003386,"patch_set_number":1,"change_status":"NEW"}],"branch":"refs/heads/master"},"8a2315ff37c6e6c8c52b3016ae55233efd1e8d17":{"kind":"REWORK","_number":3,"created":"2026-09-02 18:49:02.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/66/1003266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/3"}}},"commit":{"parents":[{"commit":"9d2b3b14f33570ec28396be9c46d0267028a143d","subject":"Fix task leak for live migration rollback case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d2b3b14f33570ec28396be9c46d0267028a143d"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-02 18:48:51.000000000","tz":120},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a2315ff37c6e6c8c52b3016ae55233efd1e8d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a2315ff37c6e6c8c52b3016ae55233efd1e8d17"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d2b3b14f33570ec28396be9c46d0267028a143d","is_merged_in_target_branch":false,"change_id":"I363c051ce112a28558214964ce276e47fe521e6c","change_number":1003386,"patch_set_number":1,"change_status":"NEW"}],"branch":"refs/heads/master"},"40d1133235d9298aaac9cc9f1957d728c925ac0e":{"kind":"REWORK","_number":4,"created":"2026-09-04 05:34:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/1003266/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/4"}}},"commit":{"parents":[{"commit":"9d2b3b14f33570ec28396be9c46d0267028a143d","subject":"Fix task leak for live migration rollback case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d2b3b14f33570ec28396be9c46d0267028a143d"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 05:34:07.000000000","tz":0},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40d1133235d9298aaac9cc9f1957d728c925ac0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40d1133235d9298aaac9cc9f1957d728c925ac0e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d2b3b14f33570ec28396be9c46d0267028a143d","is_merged_in_target_branch":false,"change_id":"I363c051ce112a28558214964ce276e47fe521e6c","change_number":1003386,"patch_set_number":1,"change_status":"NEW"}],"branch":"refs/heads/master"},"81d79f442870f112c1311514102fe5a4da3a0f6e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-04 15:30:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/1003266/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/5"}}},"commit":{"parents":[{"commit":"a154786239cbacba8bd7741fe34da72d8c3434d5","subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a154786239cbacba8bd7741fe34da72d8c3434d5"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 15:30:23.000000000","tz":0},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81d79f442870f112c1311514102fe5a4da3a0f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81d79f442870f112c1311514102fe5a4da3a0f6e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a154786239cbacba8bd7741fe34da72d8c3434d5","is_merged_in_target_branch":false,"change_id":"I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f","change_number":1003910,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9d2b3b14f33570ec28396be9c46d0267028a143d","ours":"40d1133235d9298aaac9cc9f1957d728c925ac0e","theirs":"a154786239cbacba8bd7741fe34da72d8c3434d5","merge_strategy":"recursive","contains_conflicts":false}},"095bc11dac43daf8b78fcb11b36e4f2d5670d529":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-04 16:51:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/1003266/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/6"}}},"commit":{"parents":[{"commit":"9b06c78525649dde8fc2dc6128cebe85fc627f21","subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b06c78525649dde8fc2dc6128cebe85fc627f21"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 16:51:50.000000000","tz":0},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/095bc11dac43daf8b78fcb11b36e4f2d5670d529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/095bc11dac43daf8b78fcb11b36e4f2d5670d529"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b06c78525649dde8fc2dc6128cebe85fc627f21","is_merged_in_target_branch":false,"change_id":"I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f","change_number":1003910,"patch_set_number":3,"change_status":"ABANDONED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a154786239cbacba8bd7741fe34da72d8c3434d5","ours":"81d79f442870f112c1311514102fe5a4da3a0f6e","theirs":"9b06c78525649dde8fc2dc6128cebe85fc627f21","merge_strategy":"recursive","contains_conflicts":false}},"05d9a08244c15f30ff3257a8ca40ffd07ad35d80":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-05 01:16:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/1003266/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/1003266/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/1003266/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/1003266/7"}}},"commit":{"parents":[{"commit":"5e8bcbfe9a69679c09fd141b4d24ff6ccab72304","subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e8bcbfe9a69679c09fd141b4d24ff6ccab72304"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-01 12:28:01.000000000","tz":120},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-05 01:16:19.000000000","tz":0},"subject":"[func test]Catch hanging task at graceful shutdown","message":"[func test]Catch hanging task at graceful shutdown\n\nThis patch bumpt the manager_shutdown_timeout to 300 in the functional\ntest to forces any test case that leaks a pending task to time out and\ntherefore forces us to fix them.\n\nRelated-Bug: #2166009\nChange-Id: I6f34ad32a4c5eb80131673be1572fe2fd609bcad\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05d9a08244c15f30ff3257a8ca40ffd07ad35d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05d9a08244c15f30ff3257a8ca40ffd07ad35d80"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e8bcbfe9a69679c09fd141b4d24ff6ccab72304","is_merged_in_target_branch":false,"change_id":"I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f","change_number":1003910,"patch_set_number":4,"change_status":"ABANDONED"}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9b06c78525649dde8fc2dc6128cebe85fc627f21","ours":"095bc11dac43daf8b78fcb11b36e4f2d5670d529","theirs":"5e8bcbfe9a69679c09fd141b4d24ff6ccab72304","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
