)]}'
{"id":"openstack%2Fnova~1003910","triplet_id":"openstack%2Fnova~master~I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-05 04:40:35.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-05 04:40:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f","subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","status":"ABANDONED","created":"2026-09-04 05:53:07.000000000","updated":"2026-09-05 04:40:35.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d98b96cc1a1c23094e0b245dbd913f2a6ccafb1c","_number":1003910,"virtual_id_number":1003910,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-05 03:23:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 06:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 06:48:50.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 07:22:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 09:04:27.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"987fe927bc22329ae814c77d0ac7db1b02538c6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 05:53:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a75115b4f248e8aee5683b68a73fabda78ca2a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 06:29:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/535bd15676454009b6d8ff84ff1d45f9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f777263b15f44d89ebfd085aa720372 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/794cf8f0462b4dfaabd12c5b6a40d141 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13bfe9f479f8433b9b53036aae86f0c4 : SUCCESS in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d5aa6df5858a9d103811ff53c2f14e2393de0b1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 06:48:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f .","accounts_in_message":[],"_revision_number":1},{"id":"628fa0f1fce624d37fcddbf226d1a8c48738a3df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:22:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6080788f70e4d989a175cef5411e294\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14340c0889bd48c9a953b20093eec159 : SUCCESS in 1h 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52476580daa44fd9388a50f5c98a553 : FAILURE in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/220edb50286e4bdcb4be4a9248e0489e : SUCCESS in 11m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12d3399dbd824460a17652484e6c5057 : SUCCESS in 11m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5059fa82161497ba913aaf1d29f6f23 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7afe0fd8121c45de80856865647820fb : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e15ac0be16849808b5086beea946b0c : SUCCESS in 7m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c403a675e0e04235b3ad4a328cb35411 : SUCCESS in 3m 35s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/eb0c853ce3b94e33b8b5802638665dcd : SUCCESS in 30m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fa3aaabd851b43aeb56047d017c23f13 : SUCCESS in 28m 11s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dc290c2c0f1345948d6f463b2a35c20b : FAILURE in 29m 45s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7cf4147a1aac445a9485a22b8d618879 : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8d93472728eaef774a6102ff9a9ed9df8fbb86b8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-04 09:04:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cc1ad0ee89d96083271ce3548c40f86be44ef22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 15:30:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fd47864a0fac767f27085f5ada7c72971d2ec5be","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 15:30:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90dd2bba9692834995d07ea8ab955fd6db551a9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:53:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/662402978c024097b9b1b6b35eaa6141\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/742115f39e90480fbddcf734d966307b : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2bb94f1b4f74cb5b7407a59fa01f702 : SUCCESS in 16m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05faf3f3687a4dcf9a8a1ed57919c477 : SUCCESS in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c0c77c56912a75f77983dfaf9ae83b61d2b293a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 16:26:03.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f .","accounts_in_message":[],"_revision_number":2},{"id":"0807b2fdc255786abb81bb5575426303fa2e310f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-04 16:32:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a90dad5f505ec97ac9eeae7546511a8c099526e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 16:57:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65b3d4bc17d4d9fa57a564ec365f8ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/afe9f22d53db417a958efb93a6a304b2 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12bc914d3cb44e57ae127807d1c653f1 : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/841c2f20a9924545bab33bebd197c7dd : SUCCESS in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36b7649e8b55f0ed8e8ad147631b6559dd6b8713","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 17:25:58.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f .","accounts_in_message":[],"_revision_number":3},{"id":"eb8eef4ebc532a3a2014c4f54554918cde10b2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:42:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cbe4a182ee54abdaa9428eb53ec8d85\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c957503a6f5547a29872f75e01d61386 : SUCCESS in 1h 00m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/62458b97910d4669baca64eddd5366d1 : SUCCESS in 1h 48m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/becb5c74592844768f80e9fd04feaf35 : SUCCESS in 1h 01m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fe55f330f594f5ba8b47918aa6c603b : SUCCESS in 1h 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5a368ad8d9469785aeed8d275e1c8b : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f2363db4fb54da89421577e082d0791 : SUCCESS in 9m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8eb2ae116ba4c08bca0c5e71670e0a4 : SUCCESS in 16m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e2aa7a0001140669543bfb5319347d6 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a562170cf84635a2030a7de44d2aec : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83d2d5ca4d384349a7fdc03aaff0d466 : SUCCESS in 6m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1d450a2e61464689a43dc5cc300ef7c0 : SUCCESS in 1h 05m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb10e44a7cf44d709e1ae59aebf5c905 : SUCCESS in 1h 24m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6f8e4efbb886499bbba5031dbade85a1 : SUCCESS in 52m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8842483173bb48a08126f657724b1a9d : SUCCESS in 37m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/743ce63ac0dd4e1284a67e1cba320f22 : SUCCESS in 1h 00m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5417809ca5244af6b617a0282500b2c6 : SUCCESS in 44m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e35486b6beb94eb985e8615e88339f18 : SUCCESS in 2h 02m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/68e3a101f99e41ba856402983d7e10c1 : SUCCESS in 1h 44m 21s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/05bd854d13764c1abde7318882f5b35a : SUCCESS in 48m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/169e84f91d494afb9aa4ea7580827762 : SUCCESS in 5m 06s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8333c8a326c34e4c9b780cfc92c6cbe1 : SUCCESS in 27m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b54c011c229424c8f1858e88d8c9fbf : SUCCESS in 28m 57s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/38aff126ce68482f97bdb878dd585f6e : TIMED_OUT in 2h 01m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/194f765a9ba84dc48bdc0b41c4a46f49 : SUCCESS in 1h 25m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bf0688567c040b685c4fbb79857a8c1 : SUCCESS in 42m 10s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5c3d836d37e7400eb19684deccbf6de8 : SUCCESS in 27m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/102226cd38d542df97fa5d356f3eb57a : SUCCESS in 36m 24s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3b4e892013cc4b339d701eeaa7081285 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"22efa1a5a4caf81d98ea6787bedd7188d083d380","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 01:03:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a757a392bb311fb0d351addae8bedb9de0c1b3f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 01:04:17.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"58e186355ccebdb90b80e1b7bc0d015c2a9700fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 01:39:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fe2d405dc6645188627b23ac6cfaa7d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2437c053e5a6415aa61cdab8ff1e831a : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4e8753d46054a258bac881b1ce1feea : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d801b0b58e6a4d59b9c91c7c624b2af4 : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e1457180a634498e33efd72d60d113288defd21","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 02:31:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0a5775ef71a7377ee7864ca899fda066c6b25a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 03:23:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4223f82e37cc4f819f18e7ad17a7ac2c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1525f4b711f4b279d29910b65895119 : SUCCESS in 34m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5202b030a7b14059b1365253b8a18f71 : SUCCESS in 49m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9500928ee40f42dcbb542dda9fac630c : SUCCESS in 45m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bf1aca1c20045749e7933c02d8f2e15 : SUCCESS in 58m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bd79ebd79f48f2a951b02b7860be89 : SUCCESS in 7m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3707e9689fc2477ca45a199300ac8828 : SUCCESS in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e247709d185841e8a0d1bd0ca6619d28 : SUCCESS in 9m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0579528dd4be4086ae545851b6448a4b : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b592b159b158437f919ea737f8f10bf6 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64be2641ae924cb4989d3c13ad0a6fab : SUCCESS in 8m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/496f571f659a4d61b80ee8272b6ea746 : SUCCESS in 28m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd60324e728f4883814a9e5399e5cd69 : SUCCESS in 1h 30m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2cf851020dec4f6bbb805a4d1402dfef : SUCCESS in 47m 59s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1c83c6dc1b6849b2847d88ebcafd73aa : SUCCESS in 39m 18s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/8ac77bacf6034f9d97a90f8759075f38 : SUCCESS in 1h 01m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/329de4d1e7814245bc42cf77d16047e4 : SUCCESS in 56m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1ccf422a4fd049c5ac52a4f70de1dbcc : SUCCESS in 54m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/e4639af31e464c08a8d429ef2ee41abb : SUCCESS in 2h 04m 09s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6baeb6bc7bf440fcb89a77f52ec2f8fc : SUCCESS in 29m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4aa1f88ff815445385e6f2192f1e2537 : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c0b5358446f74bfeb14352d84accb4f3 : SUCCESS in 27m 03s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fe8cad3fe6c84d3c88a16c9ba1af9163 : SUCCESS in 28m 02s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4351a8d0c6644224a26b1b5739ed3c35 : TIMED_OUT in 2h 01m 26s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/05a91c8b6b43477eb99045983a650752 : SUCCESS in 1h 27m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f51853d162e343cebbda0c809a60d44e : SUCCESS in 30m 35s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/10239c48d52b4579b79a66983ebcf885 : SUCCESS in 54m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b9861396170a40ab9896e4877d5437e1 : SUCCESS in 17m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ce4ad7f8d80b4ee1b1718c277c634167 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5b60dc64ff5f01ffd32cb45badbfe934bfc36540","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 03:41:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6ec60759400e4c0b6cc090fb9f36864c5c3fd777","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 04:40:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d98b96cc1a1c23094e0b245dbd913f2a6ccafb1c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-05 04:40:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5e8bcbfe9a69679c09fd141b4d24ff6ccab72304","revisions":{"62784716cc25eea0b32f000f2e5c87a09bd73f77":{"kind":"REWORK","_number":1,"created":"2026-09-04 05:53:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/1003910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/1003910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/1003910/1"}}},"commit":{"parents":[{"commit":"ad5e2ea26e3bd80a84e0ab205ceef4200906ded1","subject":"Stabilize _test_interface_attach_and_detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad5e2ea26e3bd80a84e0ab205ceef4200906ded1"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 05:40:46.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 05:41:08.000000000","tz":0},"subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","message":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration\n\ntest_migrate_disk_and_power_off_crash_finish_revert_migration test\nperform the cold migration and restart the service from\ndrivers.migrate_disk_and_power_off() mock function and then sleep\nfor \u00271000000\u0027 sec (this means cold migration is sleeping for this\nmany seconds). When the service is restarted, service.stop() is\ncalled, initiating a graceful shutdown. graceful shutdown waits for\nall the in-progress tasks to complete until manager_shutdown_timeout\nis reached. manager shutdown finds the resize_instance (cold migration\nwho itself started the shutdown) tasks is in-progress.\n\nBecause resize_instance task will not be completed (becuase it call\nservice restart itself), graceful shutdown takes its full time to\ncomplete and after that only service will be started again. service\nstart will call the finish_revert_migration in init_host() and set\nthe instance task_state to None.\n\nBut test expect instance task_state to reach to None in 5 sec.\n\nThe failure can be eaisly reproduced by setting the higher (35 sec\nin my local run) manager_shutdown_timeout in tests.fixture/conf.py\n(like the below commit does-\n https://review.opendev.org/c/openstack/nova/+/1003266/3/nova/tests/fixtures/conf.py)\n\nFix is to match the instance task_state wait time with the graceful\nshutdown timeout. Once we will move the tests to run with default\nmanager_shutdown_timeout (160 sec), this test executio time will be\nincreased and that time we can set manager_shutdown_timeout to some\nlower value for this test only.\n\nCloses-Bug: #2166439\n\nChange-Id: I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62784716cc25eea0b32f000f2e5c87a09bd73f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62784716cc25eea0b32f000f2e5c87a09bd73f77"}]},"branch":"refs/heads/master"},"a154786239cbacba8bd7741fe34da72d8c3434d5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-04 15:30:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/1003910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/1003910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/1003910/2"}}},"commit":{"parents":[{"commit":"9d2b3b14f33570ec28396be9c46d0267028a143d","subject":"Fix task leak for live migration rollback case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d2b3b14f33570ec28396be9c46d0267028a143d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 05:40:46.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 15:30:10.000000000","tz":0},"subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","message":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration\n\ntest_migrate_disk_and_power_off_crash_finish_revert_migration test\nperform the cold migration and restart the service from\ndrivers.migrate_disk_and_power_off() mock function and then sleep\nfor \u00271000000\u0027 sec (this means cold migration is sleeping for this\nmany seconds). When the service is restarted, service.stop() is\ncalled, initiating a graceful shutdown. graceful shutdown waits for\nall the in-progress tasks to complete until manager_shutdown_timeout\nis reached. manager shutdown finds the resize_instance (cold migration\nwho itself started the shutdown) tasks is in-progress.\n\nBecause resize_instance task will not be completed (becuase it call\nservice restart itself), graceful shutdown takes its full time to\ncomplete and after that only service will be started again. service\nstart will call the finish_revert_migration in init_host() and set\nthe instance task_state to None.\n\nBut test expect instance task_state to reach to None in 5 sec.\n\nThe failure can be eaisly reproduced by setting the higher (35 sec\nin my local run) manager_shutdown_timeout in tests.fixture/conf.py\n(like the below commit does-\n https://review.opendev.org/c/openstack/nova/+/1003266/3/nova/tests/fixtures/conf.py)\n\nFix is to match the instance task_state wait time with the graceful\nshutdown timeout. Once we will move the tests to run with default\nmanager_shutdown_timeout (160 sec), this test executio time will be\nincreased and that time we can set manager_shutdown_timeout to some\nlower value for this test only.\n\nCloses-Bug: #2166439\n\nChange-Id: I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a154786239cbacba8bd7741fe34da72d8c3434d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a154786239cbacba8bd7741fe34da72d8c3434d5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ad5e2ea26e3bd80a84e0ab205ceef4200906ded1","ours":"62784716cc25eea0b32f000f2e5c87a09bd73f77","theirs":"9d2b3b14f33570ec28396be9c46d0267028a143d","merge_strategy":"recursive","contains_conflicts":false}},"9b06c78525649dde8fc2dc6128cebe85fc627f21":{"kind":"REWORK","_number":3,"created":"2026-09-04 16:32:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/1003910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/1003910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/1003910/3"}}},"commit":{"parents":[{"commit":"9d2b3b14f33570ec28396be9c46d0267028a143d","subject":"Fix task leak for live migration rollback case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d2b3b14f33570ec28396be9c46d0267028a143d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 05:40:46.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 16:31:13.000000000","tz":0},"subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","message":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration\n\ntest_migrate_disk_and_power_off_crash_finish_revert_migration test\nperform the cold migration and restart the service from\ndrivers.migrate_disk_and_power_off() mock function and then sleep\nfor \u00271000000\u0027 sec (this means cold migration is sleeping for this\nmany seconds). When the service is restarted, service.stop() is\ncalled, initiating a graceful shutdown. graceful shutdown waits for\nall the in-progress tasks to complete until manager_shutdown_timeout\nis reached. manager shutdown finds the resize_instance (cold migration\nwho itself started the shutdown) tasks is in-progress.\n\nBecause resize_instance task will not be completed (becuase it call\nservice restart itself), graceful shutdown takes its full time to\ncomplete and after that only service will be started again. service\nstart will call the finish_revert_migration in init_host() and set\nthe instance task_state to None.\n\nBut test expect instance task_state to reach to None in 5 sec.\n\nThe failure can be eaisly reproduced by setting the higher (35 sec\nin my local run) manager_shutdown_timeout in tests.fixture/conf.py\n(like the below commit does-\n https://review.opendev.org/c/openstack/nova/+/1003266/3/nova/tests/fixtures/conf.py)\n\nFix is to match the instance task_state wait time with the graceful\nshutdown timeout. Once we will move the tests to run with default\nmanager_shutdown_timeout (160 sec), this test executio time will be\nincreased and that time we can set manager_shutdown_timeout to some\nlower value for this test only.\n\nAlso, running this test in threading mode.\nCloses-Bug: #2166439\n\nChange-Id: I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b06c78525649dde8fc2dc6128cebe85fc627f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b06c78525649dde8fc2dc6128cebe85fc627f21"}]},"branch":"refs/heads/master"},"5e8bcbfe9a69679c09fd141b4d24ff6ccab72304":{"kind":"REWORK","_number":4,"created":"2026-09-05 01:04:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/1003910/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/1003910/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/1003910/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/1003910/4"}}},"commit":{"parents":[{"commit":"9d2b3b14f33570ec28396be9c46d0267028a143d","subject":"Fix task leak for live migration rollback case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d2b3b14f33570ec28396be9c46d0267028a143d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-04 05:40:46.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-05 01:04:12.000000000","tz":0},"subject":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration","message":"Fix the test_migrate_disk_and_power_off_crash_finish_revert_migration\n\ntest_migrate_disk_and_power_off_crash_finish_revert_migration test\nperform the cold migration and restart the service from\ndrivers.migrate_disk_and_power_off() mock function and then sleep\nfor \u00271000000\u0027 sec (this means cold migration is sleeping for this\nmany seconds). When the service is restarted, service.stop() is\ncalled, initiating a graceful shutdown. graceful shutdown waits for\nall the in-progress tasks to complete until manager_shutdown_timeout\nis reached. manager shutdown finds the resize_instance (cold migration\nwho itself started the shutdown) tasks is in-progress.\n\nBecause resize_instance task will not be completed (becuase it call\nservice restart itself), graceful shutdown takes its full time to\ncomplete and after that only service will be started again. service\nstart will call the finish_revert_migration in init_host() and set\nthe instance task_state to None.\n\nBut test expect instance task_state to reach to None in 5 sec.\n\nThe failure can be eaisly reproduced by setting the higher (35 sec\nin my local run) manager_shutdown_timeout in tests.fixture/conf.py\n(like the below commit does-\n https://review.opendev.org/c/openstack/nova/+/1003266/3/nova/tests/fixtures/conf.py)\n\nFix is to match the instance task_state wait time with the graceful\nshutdown timeout. Once we will move the tests to run with default\nmanager_shutdown_timeout (160 sec), this test executio time will be\nincreased and that time we can set manager_shutdown_timeout to some\nlower value for this test only.\n\nAlso, running this test in threading mode.\nCloses-Bug: #2166439\n\nChange-Id: I0fa181a78d9a7525c9e5d1a9f88cdf2a06af959f\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e8bcbfe9a69679c09fd141b4d24ff6ccab72304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e8bcbfe9a69679c09fd141b4d24ff6ccab72304"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
