)]}'
{"id":"openstack%2Fnova~1004424","triplet_id":"openstack%2Fnova~master~Ib03ba38a32891128aff6b6cea5d3b184a6c80cdf","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2026-09-11 13:46:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib03ba38a32891128aff6b6cea5d3b184a6c80cdf","subject":"This adds an alias for Hibiscus","status":"MERGED","created":"2026-09-07 14:09:09.000000000","updated":"2026-09-11 13:47:47.000000000","submitted":"2026-09-11 13:46:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004424","meta_rev_id":"f177d9171c9f5bedb3da8977572c2b9305ce99a1","_number":1004424,"virtual_id_number":1004424,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 13:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2026-09-08 09:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-09 09:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-09 09:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 14:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 15:04:51.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 16:09:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 09:21:55.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-09 09:58:20.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-09-11 08:38:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-09-11 10:46:25.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"eaa422536e6fc37f8d048d73a98bba03674e45d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-09-07 14:09:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"336c334cabc12ce540b2a53549e0583ee5b67e60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:39:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85da9157cc8244aea73b341a6667e470\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5ccd9826a2b545f38a1cacb3383f22ff : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9826f6f9dbd94fe7a4f7df0bd8f58f72 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/828a85f341da4948a23121f0cfb26d09 : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58293020e9e1930143b960c18660f093d76b174a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 15:04:51.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib03ba38a32891128aff6b6cea5d3b184a6c80cdf .","accounts_in_message":[],"_revision_number":1},{"id":"398f0632e6b587bafa3d19245134df435f6aca4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:09:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/191060c88a684808b9e5e7a4cecd040d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8ebbda9dac54238b8da55409b39d6f7 : SUCCESS in 36m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/14ac316b855b40198183273ba3c9a457 : SUCCESS in 1h 22m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b85945099064056828db02aa814c52e : SUCCESS in 1h 00m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c34d8e8ac7a048c5a6ce3f69a6b6427f : SUCCESS in 1h 24m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47235bb16f2b488895d2c5e19dc41708 : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ce108a933534144baac6fcfd647c2fd : SUCCESS in 14m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72f719f1e015426089e559e8dc7b56da : SUCCESS in 13m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b01ffa2d2f594b3e91bdf2f7b42b1842 : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eceac553236435dbaff4e5631d6dd7a : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d20d8648a634baa8ce92eea2b8b8702 : SUCCESS in 8m 29s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c74dbce2e93e45608ea99bc8659c46ad : SUCCESS in 36m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5831caa112b342ea96dfe5626d5a0658 : SUCCESS in 1h 30m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fa57a5a6ca654dd0a8cf3b93076d034d : SUCCESS in 28m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e6b1c15248204ff1af7ab4c0526334e6 : SUCCESS in 36m 54s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/50299468706441629012a53a9907a609 : SUCCESS in 36m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8aef48a8ef26476c9e9f22925faeb8b8 : SUCCESS in 48m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/eee153f54b234295b22fb7b84869348f : SUCCESS in 1h 54m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/9f52da96020d4a6aa60c453040aa10dc : POST_FAILURE in 21m 09s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7e8ebb34136f48b99c71a25aea7082ac : SUCCESS in 46m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6bc226040f0a48b0a584c58f378de6a3 : SUCCESS in 3m 50s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c499b13695964ba2b2efdd9f647991eb : SUCCESS in 24m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/26e4b81389f54a35a4036652b8bf0a6e : SUCCESS in 29m 54s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7d9499bad2e3461c911afc7a354b50e0 : SUCCESS in 41m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/26caa3392a2c45a382843e0e1e86058c : SUCCESS in 1h 28m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cfa669f9de5444a99c6a75225e56e0f : SUCCESS in 1h 00m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/58aa1ec88ff34a379292521f5936f19b : SUCCESS in 43m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/727dbd87a8264fe88618c03093f468e2 : SUCCESS in 15m 34s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/79139365266f469bbd4770dbd227c851 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c821409e197e8b1a6ffa77a6ecfe42463472f5b6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 09:21:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12e0aad27f6903d6d86cc3860b4b5388ac0a5fbd","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-09-09 09:58:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d86b5dabd81e6c8fbc914b59be7b658268d9ff6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-09-09 09:58:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc8304bd275fb1b9627eea61686082774b14f8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 11:45:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50d195cc91bf4225a9edd97bb8b445c1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8b016cd468c4d79bea580c1c631326a : SUCCESS in 1h 09m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d3d6f96152b04abbb408ec825fb26835 : SUCCESS in 1h 39m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a856d389a4a498d943c15836b347d40 : SUCCESS in 1h 04m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eefc79a22eb2418ea3a4247798d29621 : TIMED_OUT in 1h 42m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33dec64652924e49ad7b4656a74f45f3 : SUCCESS in 9m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41e06664d17440efa453b3ce887fb2cb : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95a8b273007f4bf08be2d60d173032d9 : SUCCESS in 12m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba8503b7b1c24f6cbe88470f4c421cd8 : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70a7512dfc24612943dfb985aa342cc : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ce677873b0b4384a634d12a70e7aa59 : SUCCESS in 10m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4389dbd3f6be4f209105eb5f37667df9 : SUCCESS in 1h 22m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5618d2b2ba4e4950933a5e5b7bf20507 : SUCCESS in 1h 33m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e97add0f11684a0c9ea16479a0e3239c : SUCCESS in 49m 31s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d3f13b476a2b4cb49ac8a293c79d044c : SUCCESS in 39m 33s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/da74a195621b4895af75c37a7b71a68e : SUCCESS in 43m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/58c84627fbef442cb195c5ca3c9e5853 : SUCCESS in 32m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f5d14545040d46dd89d91524b4d37456 : SUCCESS in 48m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/a53f563e41454c16afc457335a397ce2 : SUCCESS in 1h 34m 02s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/fa79f40f9f7746559774ab0c625656c7 : SUCCESS in 29m 48s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9c52a05902f4469a9c0d578170574aa0 : SUCCESS in 5m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/caf382fb9df14ebe87d9bfd829062503 : SUCCESS in 28m 53s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f78a71e7ca2141e69a8170f5aeb2d31c : SUCCESS in 31m 10s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9b18bc3cba3c429394ae4e290d61b63d : SUCCESS in 32m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/87db7acf1b3347a78f863fa94e96fa02 : SUCCESS in 1h 20m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cba1c5b3fdb748db87e8e48835d6a537 : SUCCESS in 1h 00m 03s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d1fab93e3743416ea191a8dc5cd316f3 : SUCCESS in 28m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d482646641ff40ecb15c12cb072a86a5 : SUCCESS in 24m 08s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1b234ed61a5e44d9b38efa896ddd1a17 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5c6997a4e2349ffccfcfd493f46f0c7349494886","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-09 14:05:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6ac26220a70713f46ee22921ae063d51602f84d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:19:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17fe6efdc8948058a0037c2a2babb78\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9f195f66bf94202a42588a7c8225bef : SUCCESS in 21m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8a52550f1374b42b1970ded284b223b : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0deee1f50c7f49b6b0cb6406fa02e733 : SUCCESS in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c64d45baf1baaa71d3ca106dd524ce1e9fd2daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:49:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70e7649de7b14a19991b46eb84c07da8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a2565c102454efe9fa4facd48f0ed5d : SUCCESS in 1h 25m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2cd147431fb74e9bafd5e0f261060bd1 : SUCCESS in 1h 45m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/435af1a430b84976bf91b0913f79b6d8 : SUCCESS in 1h 02m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1329adf595244e878f51429a6285f7f9 : SUCCESS in 1h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607e0fbee74546718a5350485943ebc0 : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e41c7f2e30d4b16bb1c65ba196e31c9 : SUCCESS in 19m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c022de4a9ff4b159be065de7f7e0675 : SUCCESS in 13m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9be5147312a54aa6b17907f4a5c845f5 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7befc0cefb3b48a9a964c2c107bdd8f7 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da223f8c22fe45f984428283f85889ba : SUCCESS in 8m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/38853ae0d2f944418ccf8a162f37217b : SUCCESS in 1h 18m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d6fb1d0d3a4e405b814ed96ef6067c3d : SUCCESS in 1h 37m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d3773fce384f4c96930703dbf7be3722 : SUCCESS in 52m 54s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0d72ae7d5cb34712b614fee06070fe74 : SUCCESS in 41m 09s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ecb46a6b9e924e69b933a0cbb1b3b937 : SUCCESS in 1h 02m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2bfda0a289104e109c6ea6de92e71b44 : SUCCESS in 1h 01m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/19c0145ef7454341957509c8c86b3b16 : SUCCESS in 1h 59m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/01b6610fe4e04486994d52e96b1fb63a : SUCCESS in 2h 05m 17s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/66ae43af7a34455a83485e0953e00994 : SUCCESS in 51m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cebdf4f761f5402ba4f9476dd2a92238 : SUCCESS in 6m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bdebf4e132474028b0ef0013001751eb : TIMED_OUT in 1h 01m 15s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7d4be4cb92fc47efbda8e00e9f22d12d : SUCCESS in 27m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5037de2520de495aa10c4c3e15f6be28 : SUCCESS in 38m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/67ede83e36844ab7a46d4c4a1e3b4914 : SUCCESS in 1h 32m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/569a06e8b7814fb8964500fc648cd09f : SUCCESS in 1h 04m 31s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/7474f145184e4833b83bff3b75d8d9e7 : SUCCESS in 59m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e116b59f510a48c888eb854c3fe903b4 : SUCCESS in 40m 10s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cb2e478474c249e982688e298e1bdab5 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b320269cca04a6c8b13e879e3c90f15361124b7f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-10 14:37:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7eb1ea9e50fe75164e7fea2df878a7be6fc6060e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:17:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9be3fe60b474ea5a22ab2545e7d131e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/811c3f92e9dc42d4b4211abc8ef6d814 : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fadd41f2e0e245f7b089218762dc4359 : FAILURE in 20m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c63023563dc4479f9b324d0a5a41562e : SUCCESS in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13c84ae9d440abbcec7eb9d14d52b6dcb4ae81a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:44:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a5a866aa1ed4b5fb8b8cbd017dc6c37\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b254e747946f488eb56709ff16c7c2d7 : SUCCESS in 56m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/57f57c76b2224fb6a6609c2000eeb5ca : SUCCESS in 1h 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce5b73ec353e4f97aa154ba75ff62d84 : SUCCESS in 48m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4d3f118f17c405b9769018697a16f33 : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ff0a992c3441bfba46405514754f23 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c24d521eea34897b4b55407323eaea8 : SUCCESS in 9m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a37302854ef94c94a9c6907988a09ccc : SUCCESS in 15m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd96b2ba129d4608abb59fc09a56b59a : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c407a740334cc7843e3c7b1f4e494d : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98285c4a090242faac783a9e8a203232 : SUCCESS in 9m 20s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/44b1f3d21fab42a89351b100b4148cce : SUCCESS in 35m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1e36e2aea2ce4151b403be31535c9b1f : SUCCESS in 1h 34m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/45632b5cfa8d431db06fcb42ad5bdf0f : SUCCESS in 53m 08s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e2b987d22e94497fa15c43933a68d135 : SUCCESS in 41m 44s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e87adff363b24a9db41fe15b6238c4ba : SUCCESS in 38m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a871090497bb488eaaa927be4c3edce0 : SUCCESS in 25m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/014475de9578411eb29d04c883994d55 : FAILURE in 1h 54m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/958dfdbf246e43a3861bbfbbf8294efb : SUCCESS in 55m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/671948442f864d03ade01a2adc167c65 : SUCCESS in 56m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2d759d8f7e264101a73302200d39d71f : SUCCESS in 5m 47s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2ac3cecb680f4913bfef6405816e2dc8 : SUCCESS in 36m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3e3edea58db84fb6b8636efad0d23e7b : SUCCESS in 28m 05s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/31a59590a3d741a89885044d5893baf8 : TIMED_OUT in 2h 01m 38s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5a702a992c544a28fb41dacd0238929 : SUCCESS in 1h 38m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0d1ae0cb51d4d8c99aa6e2626b9355f : SUCCESS in 45m 25s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5b91333b58274c3580e782a735089bfa : SUCCESS in 51m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/42c4518d2e314c89bc5b391c1a2cb368 : SUCCESS in 34m 50s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4e593a85f78f4cc8a35b83be90fcd97d : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4ad20cf93502ae0368a1ad48323867f3c05f059c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-11 08:38:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7056ac9169704aa9e4e875ad2fff7a55d6ce41c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 09:20:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea9a4cd75c7b414aafc1820508884392\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/378d40c5f5ab4a61b36b72be410ff3be : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a030afdc3a54a5eb64224b720417c18 : SUCCESS in 21m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1956180fdfa649de96af611ad89cde59 : SUCCESS in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"954d049f7615176b8324177aac5955cc32517c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:46:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3d67f5371614bcda2bc747b8cfba15e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ddf8b2a57134dada8cc95f0423ae1bd : SUCCESS in 1h 04m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/37f670619ca94f50b1a554aa07c4f973 : SUCCESS in 1h 45m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2d86ab0e8a94d439f49d2aa4f127e75 : SUCCESS in 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fdabcd8ce2b4d23bc12e9f66beb61bc : SUCCESS in 53m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72e4bd598cc4d3f85be6bc165c5d196 : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6b9dee6fa5d4552ad52b39c5e99e648 : SUCCESS in 14m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/682b63cbb4a14c3baa8a7a23e9dc18b5 : SUCCESS in 8m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/300d000e874b42219341a2058e985d6a : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a48fb1f36a4540a736ff7979a1a4a3 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b26d71805ffd49aeab8c89089c276c5e : SUCCESS in 7m 15s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0a22483396484d12890a7bcdf9bc2588 : SUCCESS in 46m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a294415f860472a9c9cf63d05c7d55c : SUCCESS in 1h 37m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/65efa6ddf1c245029785552aa428045c : SUCCESS in 50m 59s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ec5ac891753b4daea145dd5958e7462b : SUCCESS in 37m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/8582172f27254e84b97b24ec897c90ec : SUCCESS in 1h 02m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b5156b883b3b4e9c88b6f1bf402877e4 : SUCCESS in 49m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/401c2da5b6a94bb38521d721a81f597d : SUCCESS in 1h 27m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/fefb05d5379f498092bb9f4c0e5e2c53 : SUCCESS in 1h 58m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/49d0ab59af2c48a8ad7b5e19177b9a3f : SUCCESS in 47m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ac508e6bd5954e39b535d618886389c7 : SUCCESS in 5m 03s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/960a07bf78eb4e29911c0368dc0258db : SUCCESS in 29m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e54f7344f9be4e40b82ed29b7477b50c : SUCCESS in 32m 48s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f2b8308371ee4293a1922edbd1b9a782 : SUCCESS in 35m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dde8809cfcea470f9206dd886bcf28e0 : SUCCESS in 1h 02m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b58709f632a4e9cb5027de27f6f8ef9 : SUCCESS in 36m 58s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2e4d5a71119d4bec947d11e849c957ff : SUCCESS in 39m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/474d48bd679745db98892974f4bb2988 : SUCCESS in 15m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/975b9ace95c9494ba45818bddbf842e2 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3f3489f24b34d334fd159a9192ea08359ccc46c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:47:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f276d995e2831086f0f61f1eb77063e76034227","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-11 10:55:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"5d9af992b3786b61247805a476bc34c6c338a195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:46:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9bc485651244e4ca0722a93ae21dcc9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e3cde9df2f24f3ba037c7a5050d274c : SUCCESS in 51m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/831eb753c952499491a847afdac04435 : SUCCESS in 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8964c4937024fefa43b767306679200 : SUCCESS in 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103744b9295e40c1a572da53061095ca : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df2a4211667a495bb6fa0af90f09dcc1 : SUCCESS in 9m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e417f6d550e04f0a99f3accb1fb72300 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9536f4a1c8a24b1da2241afb0e054122 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c78e005874a4192be9a51729a32b25b : SUCCESS in 4m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ad1e30ab41b3486797a572fa9b883a1d : SUCCESS in 25m 16s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1cbef6f3fa8849b2bd44ed72930c75ae : SUCCESS in 27m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/affccd571d9e4e8280a9a06465a6b288 : SUCCESS in 22m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3ee42794d9294e888a2ed0a4f4bafb83 : SUCCESS in 28m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c1e2c388381a479880db541a70464775 : SUCCESS in 1h 56m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/94b2b53d8ab2408996444e6a46c9c76e : SUCCESS in 1h 32m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e46e577e2bc348df9d9abdff58a095c3 : SUCCESS in 6m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5462cd46f5a54a048f98d855c3ae353f : SUCCESS in 1h 08m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/463202fac7fc4fc6bf3a9ed990821a9c : SUCCESS in 1h 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0530a96282b942fdb97ace487db5d8ac : SUCCESS in 37m 33s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0a738dce62724fb0a3b23a0b6e8b9847 : SUCCESS in 10m 35s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/97f961fea9be4892ba4bc22e79beed51 : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fed76fde2a019234cdeebf841915f4493784d162","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:46:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f177d9171c9f5bedb3da8977572c2b9305ce99a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:47:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92e2d809d7e34e3791d64b7b44a33fb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62e0da1a8e64c96a8384ab0b3d00266 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13f868ef916f202ced9c596f33aec2087aa1dd45","revisions":{"13f868ef916f202ced9c596f33aec2087aa1dd45":{"kind":"REWORK","_number":1,"created":"2026-09-07 14:09:09.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/24/1004424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/1004424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/1004424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/1004424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/1004424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/1004424/1"}}},"commit":{"parents":[{"commit":"1dc44b5988f75683564bc4e765be1f84c3d75dd4","subject":"Only load same-host instances when checking share usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dc44b5988f75683564bc4e765be1f84c3d75dd4"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-09-07 14:07:53.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-09-07 14:08:57.000000000","tz":120},"subject":"This adds an alias for Hibiscus","message":"This adds an alias for Hibiscus\n\nChange-Id: Ib03ba38a32891128aff6b6cea5d3b184a6c80cdf\nSigned-off-by: René Ribaud \u003crene.ribaud@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13f868ef916f202ced9c596f33aec2087aa1dd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13f868ef916f202ced9c596f33aec2087aa1dd45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
