)]}'
{"id":"openstack%2Fnova~1004430","triplet_id":"openstack%2Fnova~master~Idd90094fd766b6d8f1bc6173600fef14e583a6f4","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2026-09-12 19:56:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd90094fd766b6d8f1bc6173600fef14e583a6f4","subject":"Add service version for Hibiscus","status":"MERGED","created":"2026-09-07 14:24:14.000000000","updated":"2026-09-12 19:57:15.000000000","submitted":"2026-09-12 19:56:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004430","meta_rev_id":"fd1c64d1c5650641a80eb66e842fa74d726bab99","_number":1004430,"virtual_id_number":1004430,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-12 19:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2026-09-08 09:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-09 09:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-09 09:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 14:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 15:16:54.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 16:32:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 09:24:43.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-09 09:59:04.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-09-11 18:30:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-09-11 20:32:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"dfc940725e51edcac5cda855b5d57ea2cab7c7f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-09-07 14:24:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dad8a2198634996fcf0dd216b043a21d160aee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75734f13348c49e6b6cf6e38d31bca3c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe648f9367584cea8ed21ec6c100799a : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dfe09e834cc246bdbb7577f10c29321e : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff2dc9212a7445449a4c0d4e942ca83c : SUCCESS in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69f1fabf6cf4e11e9273a1fe46273a8ec562047b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-07 15:16:54.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Idd90094fd766b6d8f1bc6173600fef14e583a6f4 .","accounts_in_message":[],"_revision_number":1},{"id":"abb20f74937403e1d24a18502453b5113f5ab1ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:32:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bccb8751d10f40c18abe9fc87e18c789\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15f4e0f0891a42df902378ef0c6ad94c : SUCCESS in 1h 04m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0d1046b150b943c790d0ad3af69fb7b8 : SUCCESS in 1h 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/700d15035e1a4481b26d9670d3c46c04 : SUCCESS in 1h 02m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671f82eca42c47e89edabe1056a372f7 : SUCCESS in 56m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cc4ca01c0949cb8cafdb3c7019d8fd : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/974cfaf0c6d642d1b791efcec74fcb59 : SUCCESS in 10m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8eacd0d94374c889602959f5e10afb5 : SUCCESS in 12m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93e056009dc74b2db9a395b860250296 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8513ccce01486a82b271b68de60ac8 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286fe719c80446b1b1d4e04bb9807abd : SUCCESS in 7m 39s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4dd19eaac0284c47b8812359aecc088d : SUCCESS in 1h 12m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77dd589707f44d6d93f8a685a38d54bc : SUCCESS in 1h 29m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/43dca11f56d746909bddc8b905d962d7 : SUCCESS in 56m 00s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/17c21575a2d64315bfb1d81b28cf6511 : SUCCESS in 1h 00m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5e3a75ce8b524789818c8ab172b62ecc : SUCCESS in 49m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/25947585de17408fa6cf4abbe2a6ac72 : SUCCESS in 49m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cee10dedf70c4833b165ffae4a5373bb : SUCCESS in 56m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/99ca9a00434a49888cf61d8f060069d1 : SUCCESS in 1h 59m 18s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/805111870efb48b2988c14c6ca21b67c : SUCCESS in 46m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/635568bb91884b33add44cb448c1f19d : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ca03fa486ec84893a8462c84f2c4907a : SUCCESS in 23m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f6351df431564ae68775ed49284ee723 : SUCCESS in 28m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/27f02d50be8649f7be5d7c76c5087b3e : SUCCESS in 34m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c060ca044ba48c58c0b4d09b978820e : SUCCESS in 1h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93152a7d54414fbe8b125ec10a03b5e4 : FAILURE in 7m 43s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/09368097d7794a2c9dfd00095399780b : SUCCESS in 31m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/117913dba672486fb5b2cf5f327d7b03 : SUCCESS in 19m 41s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/61ca27864e92409c9f129ebad0070d3d : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d084527f6ff11d1bd77c03ab66aab7d405d50cdd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 09:24:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f3bec7adb10ed244e1358a401790d600ec474955","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-09-09 09:59:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7bf2fb78fbdb22eb9f1e3a057fa11023301815e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:07:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b29590e096040c3abdd47024db78786\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82cee9ae03774ac5813fdd1c085819ce : SUCCESS in 1h 09m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2e1b0a7aaf46475288c797de45bdfb54 : SUCCESS in 1h 05m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/210f1b99265a43c482d8de10fa99b446 : SUCCESS in 47m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe90b9d98aea44af82934195ebee50e8 : TIMED_OUT in 1h 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8703f775e26d42af80c0816a475f5e42 : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/186895c8408d4edfafbfe40a09964cba : SUCCESS in 13m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e90942fba7642b3b83dbc2445d62626 : SUCCESS in 11m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ab0d889a721848d5ae87fa48b1869e60 : SUCCESS in 17m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b43021eef0b47209f0151e74acb9dc3 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/290cfed86513491397d02d2d0d3de3d5 : SUCCESS in 9m 52s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4c696443b7d043ea868c49699cf7c22c : SUCCESS in 1h 10m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe0875497b824d6ab5052d0848c94923 : SUCCESS in 1h 03m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/13a550c1e8bf463a916d8d561703fe05 : SUCCESS in 52m 14s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6303dee819374875acc9994c11e2e914 : SUCCESS in 37m 46s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/60377157c52e41c3a994d6e4b8d5c587 : SUCCESS in 1h 03m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/62c5570717354359961fce7bc9d12a05 : SUCCESS in 48m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/55663515998b4f8a94be31b6c71eaff2 : SUCCESS in 1h 32m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/9157b89f23a6451584faf13b32533dd2 : SUCCESS in 1h 19m 55s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0fd1ebc5981b48d8882043df4742d13e : SUCCESS in 45m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/54318538d75646ed8a574a0c655e53f1 : SUCCESS in 6m 55s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4027317183944fc298290f3b92478f16 : SUCCESS in 28m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/336ad3a3b46e4422971d10c56e317fe7 : SUCCESS in 25m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1146b631f1b340c2a81f04097eeb49a5 : TIMED_OUT in 2h 01m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0013608e7fb84329907b0a29ed1f723d : SUCCESS in 1h 37m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89d505d683cd4bb5ac9ca90e409fc92a : SUCCESS in 1h 09m 11s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/63699ef81a3d420ab51009beb642110c : SUCCESS in 52m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9f40883e2e664a2fa408481cc67ece44 : SUCCESS in 21m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1003dbf7994a45c7a82f62fa379eda39 : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e87e5bad3564d339644d6b2954a0e5b927ae7e44","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-09 14:05:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a4e58df24324a4106778690bd90c617a09051a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:23:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc7f34be8ff410b818ddd7b477d42c8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a019aedcfb09405b8621b08f287c492c : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae31b7f85c2c42e7b90df2dc5e915488 : SUCCESS in 24m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e6c5631cc9674116bbacb5c1ef2f8ba5 : SUCCESS in 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5c2a81ebeffbd98ace517f944b45297bc12ac82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:45:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75406a860eb14fe1847d7dafe3b45397\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c5c07f557c7f4484985b6d24d0fc537d : SUCCESS in 1h 21m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/82cfcce8cf6f4c39ad5df6a411405ffa : SUCCESS in 1h 01m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0459495c8eb94c1a935da0d4988b2afe : SUCCESS in 1h 02m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1443b47c3636401d9fcf3fe608bb7e7b : TIMED_OUT in 1h 41m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e10fda198584ca39b22f6ada4e713ab : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03254494c8a6495badc79975666c029d : SUCCESS in 17m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11c418d84f254649991818a19e3c32b9 : SUCCESS in 13m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02b2ae0f76e54f8a801104ff1dd95739 : SUCCESS in 18m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d285eefa054b6b8f07cf8f021b54bf : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfb530ae31d449789cf2b18d9ad283fd : SUCCESS in 5m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e4b6eb3e07b2480a81501969946846c2 : SUCCESS in 1h 29m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91beaa5570cf4ea988379a956c8b0ad4 : SUCCESS in 1h 26m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a0aab6a351ec41b78996a031b163aef5 : SUCCESS in 35m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/808ed0dda1054068b5fc29e128871ea2 : SUCCESS in 40m 58s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/64af3dbb4f1644ad8610d9b0ac55fd74 : SUCCESS in 59m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9f3bca005aee43cd9ba1471eb23ba350 : SUCCESS in 1h 03m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d531136d5b204b6183f1ac6b39a5cf58 : SUCCESS in 1h 48m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/c7a31da7b5ec486eabda2b8cc485110a : SUCCESS in 2h 21m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/414221f9c4a34017b1b1b438fddfb320 : SUCCESS in 50m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/09bbe7d861c54e618632807f268cd9e5 : SUCCESS in 3m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f602f95e5e7f45428962e597c195d371 : SUCCESS in 28m 45s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b21b9fcc5c4147f9b03700a672e8cc60 : SUCCESS in 35m 22s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/965abc3a8fa543599b3e02efe9d79940 : SUCCESS in 42m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4fc12fafaf64fa08d94e17a5340563f : SUCCESS in 1h 09m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fbfb924c42e43ba8d80ae92861d70ab : SUCCESS in 56m 20s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ac3984c53bcc417986b09a7967f98f98 : SUCCESS in 31m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9a5e970fd59449bc8d0968c2ecf81e6e : SUCCESS in 33m 45s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/150e8571d08f4078afd53d9cb710d55f : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5bfe8e73e82004ef007b9af3f55b23f1a3bb444d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-10 14:38:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d10e5ddedcaf982fff95bc36996a82fda0b5a2fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:40:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77ccf428cf3045cb8c6e0eaa57eb8f7b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ffbfe26f215645cc9e9eecf19f1badf4 : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3cefeb345d234c58a6351f3dee59befd : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a6f3321ce45e47ef932d45c0e5e6a5db : SUCCESS in 35m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"762d33553f240563c71610b7ca9e7c0fbc6307de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:48:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f957f342b1e4dbb8e1e2933b1f6b27e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f40e5f40af64fe284e3557b05e7d056 : SUCCESS in 36m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8f60be63c8c545bab7451e46b258697e : SUCCESS in 1h 40m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30b7f59210224d999cd305a873a46820 : SUCCESS in 1h 00m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b93098eba9444aba602b3242d51aa8e : SUCCESS in 1h 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5151697347824614b79042538ed42a25 : SUCCESS in 8m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0be5e3754fa048d58c390a11d5c99f26 : SUCCESS in 14m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08f414acb0f8404d80cabe18a3e29e68 : SUCCESS in 19m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c15bdeb66664c7ca2563233abbc316c : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990a604a395f47eeaf6cf3c06229df6f : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/547c980d0e9b4abeb7efed0cafd143b8 : SUCCESS in 8m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/17a4bb2dfb8a465abe035dee5150365f : SUCCESS in 1h 04m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ae51e996a5344fcda882c0473eda3f07 : SUCCESS in 1h 32m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fd2e9307c1b14cb8a1ba41dfb1279310 : SUCCESS in 55m 28s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f402b2c27dde46f6a46d4daba21c757f : SUCCESS in 38m 09s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3511954cfba64b0897da50af362c364c : SUCCESS in 1h 02m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ac1558077f9e486097b1ace646856c90 : SUCCESS in 50m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4515e69e06514786844f40ea312ff911 : SUCCESS in 2h 03m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/9f5d33b4ec9f442d917b4f91768b0716 : SUCCESS in 2h 03m 44s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/964a18ba307043548b4ec1657584b0f3 : SUCCESS in 27m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/80f4cdce78fb4246966c9b439397e4ba : SUCCESS in 5m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/20c25da616ec41b4a3c806f458a178a3 : TIMED_OUT in 1h 01m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3f39b317c91e4bc1acb4fbd17e374fd0 : SUCCESS in 27m 47s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b2dcdc6a08cd4e3c86e0d188b0dcd5b7 : SUCCESS in 32m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f8c7046d3254eccaa53ee6295c8b212 : SUCCESS in 1h 18m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e024dbf2335148c1a4530e71fed5fdb5 : SUCCESS in 38m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9249051ac2864bce9a9c9a8fc8e4afc7 : SUCCESS in 29m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b1ff492153704334ab67faacb023efba : SUCCESS in 34m 34s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8b2941e631944e4bbaa4e9be48599248 : SUCCESS in 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"dd24440d718f602a3451cbefcff35a91c78515c6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-11 08:44:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b6f559393f405b55137427f033dd06d75b53a8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 09:23:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9cc64ab0cd54cf79ccdb484f184f78e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc20d8ad13884d04ab97793a0341ecf1 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab4cfdba39f54fc6b47b52ea37678b21 : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/757856ce5506465597069cb87a43808c : SUCCESS in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25121450ff8b002415435aa4b1c309c80819aa6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:53:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e370871c78b747d193fd060321162f67\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d50951ce4df54613a8a008ff157c5486 : SUCCESS in 1h 04m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7f3f55b0b1a9448b9e5f9546d0132cd1 : SUCCESS in 1h 39m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/228f28db35364c379dc1c6db62237c4c : SUCCESS in 56m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f5d6a6cdba4413b85c1396e23ca5c4a : TIMED_OUT in 1h 42m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1efaabc4c81d450f8fe214cabf89ff14 : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76d392988b8b4af1ae79b33023f85798 : SUCCESS in 14m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c550adad7c4462c8cbd5426828d467e : SUCCESS in 15m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/651cccb75a1f4e9aa8620eac51a20fc9 : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26eb558ec8be43f0814fe920ed71cc2d : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebd0004e6e274db7ad675662b39c0518 : SUCCESS in 8m 08s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9b9216b009184c2488fea75f5d9b2718 : SUCCESS in 1h 16m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eebe0af1ec29493a88c822f4f0d79eba : SUCCESS in 1h 17m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b612899d2f664d1f89b755e845b6f2e0 : SUCCESS in 52m 42s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/82d7b82db8a14d5184f1353567cf9eb7 : SUCCESS in 37m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4e7f1b5db83649c994a8697b92525220 : SUCCESS in 49m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b06f4e1403ae43699a8e7a6781b36735 : SUCCESS in 43m 44s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ad99989053c44b358a58fd08eee59f04 : SUCCESS in 2h 02m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/e5414c605de34b33ab9fa991e9025c10 : SUCCESS in 2h 00m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/dbb684ca0a14486a9977dcb6c2dbe37d : SUCCESS in 40m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7b7034bf5f5f43048c3c946a009deda9 : SUCCESS in 5m 01s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2ec04bda9b5b4b6ca2803cca3a1dc33b : SUCCESS in 28m 49s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b3d7f9893fa14d8498c0966074b2e1ed : SUCCESS in 26m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9a24a7aa038d481eaffb5104e73b4665 : SUCCESS in 29m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d098bc4dd78e4ecba973e7d35a075d39 : SUCCESS in 1h 22m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca2bcfce3acc45098f92547dc6c37621 : SUCCESS in 56m 43s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c1d10b0bf60840bc81b751faa285b045 : SUCCESS in 28m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4b6665bd10584be29fc7b93113ebe2e2 : SUCCESS in 27m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f7ac2c5041494d0fa1e04e51b5309a5c : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"630c71331a62f41dc602877e4ab5a65a9e03abb7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-11 10:59:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ba18feb96618be18f1e837b2891f50600d2b4d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 11:22:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f7368d539d4dd4a59ab1f21ad48fcf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f7bca53165694019a58bb31ab5fb2d9a : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29084a6ca20b4ed9b58f95f904b78889 : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dd77dd77b2f94af7bf3f679ff08e54b8 : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57d1539cfca4e01b80b03eb1a8a038998656527d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:10:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1c0fd618b8e4374bda837d6f8275560\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/976545130f6a4aca945b38c5149a8ac8 : SUCCESS in 59m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3825393f647446daba6c202ff17f56ba : SUCCESS in 1h 35m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/827ee32bcbee44d59e14bb4b992b7db5 : SUCCESS in 58m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b6543c64c1b43859b068de02153249a : TIMED_OUT in 1h 41m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c48c60383a4a68816485eea4a45495 : SUCCESS in 7m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1e06e2b856045ab943c81f6c0b254b7 : SUCCESS in 13m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f304ef2ea9c44a9ea791c88f7f4a4098 : SUCCESS in 13m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/28a558306ea446739a1f226e8311559c : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0dbf71316448e18d7ffc3fd0f2fd1b : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f0dd1bffc8c450198b768f71fe33742 : SUCCESS in 9m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/41fcc48cb488467481442d3f6bec081d : SUCCESS in 1h 07m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f108d2e9ccf47a8930e7728bd97447a : SUCCESS in 1h 34m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/705ba6bedc244c1e9ee11dcc833a29f1 : SUCCESS in 53m 58s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/44177ffcb5724923b551989c7854cc96 : SUCCESS in 38m 05s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/71ebf14018a34fb1af522fac5f5f5b84 : SUCCESS in 1h 03m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/26c5d99a8386458f906d4a2f44d5f7f5 : SUCCESS in 48m 00s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/45b59879aa8b401a9beb6456c5c96229 : SUCCESS in 2h 01m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/85baeddd7f054000a7b346642f877492 : POST_FAILURE in 1h 55m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a8691f909808438c95161634689d64b3 : SUCCESS in 48m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dcc3138381614210b74b212026d778c1 : SUCCESS in 5m 12s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b0b68cc7d9c64850b953745dcb0d3bae : SUCCESS in 33m 17s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3511d2abf70a4b4ea6a8366ce0630cf6 : SUCCESS in 30m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4cf0071556be42459c17d47419ddc9f2 : TIMED_OUT in 2h 01m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a2483a1e2514f42974b4afb225d7ac9 : SUCCESS in 1h 31m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1a90b48524f4aaaa5817ec6eb95fe5d : SUCCESS in 53m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/cc8bae6782d8475d8f7be90f067f97f5 : SUCCESS in 1h 04m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5f47ac9c00fd4b9db37dbacec17632f2 : SUCCESS in 21m 52s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/231d4a06dc9f4d1dab81aa35c8344d66 : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"84a3563948bf3bbbf17d04f026995ede69db8946","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-11 13:17:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eaa69413f536d5d6b4be0b87805d8e0e74dfeaad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:40:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a4c04c99db4393bfe7210c7a642b70\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e7ef8bf17c6423ba575aeb9e7303442 : SUCCESS in 20m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5677417cf9eb42369920e4de32c148d4 : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/db4e12d3e4944168b8902c477c8ec0dd : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad85da0aef0cf373c2352183ebba0545f040d4b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:22:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/833c68f597144908b3eecfc5f1377460\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb9a09716d8d489ba2435f5f0bd45f6d : SUCCESS in 57m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/259ee389ba754162b6986ac8a67eab72 : SUCCESS in 52m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d578d3aa8e054f1e831d8fc69e722d9e : SUCCESS in 59m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ff87f15765b4fa08cb792a2e517a15e : TIMED_OUT in 1h 41m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbed054ce4024d84b727dfd7287f4881 : SUCCESS in 8m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be3550d8a50a45bdb7711140521acb93 : SUCCESS in 15m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f9ebf07da554749a2d751dabc04d230 : SUCCESS in 13m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c8664db5b4c47d5bf71dea908948dba : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372531e219f04acfa73a9b28ffc43ecf : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c5a53044ef434b883957a9867f0bb8 : SUCCESS in 8m 43s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b9f8b1345033416eb34ec775d535afb8 : SUCCESS in 1h 04m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d666034c893437ea5898a7c0a72be12 : SUCCESS in 1h 29m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/191399ad68b541f9ac5df40813938767 : SUCCESS in 34m 47s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/410f962722f24f60af4420a1078e5b2e : TIMED_OUT in 3h 03m 44s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ae86e0cb95c94154bd63b961dc35de23 : SUCCESS in 38m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/335eed0e8a15416c972cfe5930e04c13 : SUCCESS in 42m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0ad8649115b44b64b68c71e98530c52a : SUCCESS in 1h 52m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/cf2fe93b30d149ae9501310b2e8487ee : SUCCESS in 1h 49m 52s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/25caeb304f9f4e47992a2f668383863b : SUCCESS in 29m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5b0aa42e7fb54b639b9b4c71ffcf3539 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8c3df6e74dd34779aea187e32d229d9e : SUCCESS in 24m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8b99c5b25051456a87deedf0541a14ab : SUCCESS in 26m 47s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9e8a6b2343704fdda2297b1c127e35fa : SUCCESS in 30m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4474dcba35ea4faaa5b856ae8758dca2 : SUCCESS in 1h 14m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/231dc039478a4d359684ba603db51ceb : SUCCESS in 27m 40s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d15b6edb10724b66b02247d12c39fb44 : SUCCESS in 26m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1f4c741923a34496bf7153edcd6a326c : SUCCESS in 34m 38s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/08a910da05034b1ea682a6dd0db91830 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4cb789c1d7e6ee0df0fa5d9d26b65f845a1b3551","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-09-11 16:26:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa8946d87f116dfd1eb723f80302643c93841779","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:48:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f218921df84dbc8db7450b4f961a2e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3a5e808d5c24522bcf56e75cbd0d1c8 : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9ddf9072ecd470cac357ba75da93f35 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fc384f158ede44dbb1de5e056464e9fd : SUCCESS in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adca1e17e1f9ed5352074a009b3b82364eaf7625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:28:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/504af480c83941ac85643d032dfd6ef6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee59d71b1ca741768c25f8b638263d67 : SUCCESS in 33m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d5d310febd814f2fb9eea06c3120ef35 : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66dd2a8470ab4d0baa6b79ae4d046822 : SUCCESS in 55m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/989a9fbb89fa4dcf93e248866f6dc827 : TIMED_OUT in 1h 41m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd5ae87dbc74c25866d52705d513ca2 : SUCCESS in 7m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c4e5922b32e4396b7866496e2b9c40e : SUCCESS in 12m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/713978485e9249c883677654f865345d : SUCCESS in 8m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bacb8cd092a44a989c674b56b0bffb53 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becf10e9ea604f4292b35ffb07d6b419 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8886bcfbcd04165975c2ff532131135 : SUCCESS in 7m 50s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7d239699b99643098b9fcf1446a4f9ca : SUCCESS in 56m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5c118ee734a41e08d0bd1626f45044e : SUCCESS in 49m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/24faf50b67d9469b8d5ec44f4d3033e4 : SUCCESS in 26m 19s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/415d36e9bddd4b03bf18802b7203f90a : SUCCESS in 36m 54s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6ee3935d0d7a42f9b8fc6ee604246a7e : SUCCESS in 38m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a85baa5961e54d449e99604b47e06740 : SUCCESS in 27m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/402deea639e743358c63191b74a92e50 : SUCCESS in 1h 55m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/7de2471148404b26abdb3b6d350d8cc8 : SUCCESS in 1h 05m 32s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d726797c0c444d80b5f25ffd8f5dd2b0 : SUCCESS in 49m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d8b4ca586f6c427496589b36852ff0da : SUCCESS in 5m 43s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b47061ca66f9480c951f610e5b26bf67 : SUCCESS in 26m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2565475045e74b2292a935e3ab78c4ae : SUCCESS in 22m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c3b2feef1e134538b1b2522b24c39e85 : SUCCESS in 30m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f44ce2776a484db39d71d73ec083578e : SUCCESS in 1h 12m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/930d910c2e7342bf91ac319fdd10db28 : SUCCESS in 59m 43s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/628dd2b9d430427da98ad25f2687205f : SUCCESS in 28m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e4fc13911e81414887e79b736d29c654 : SUCCESS in 34m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fbd83623475d4ad5bfcc2586c7dcce53 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dce300e69ecbd5f0371a3613e25fad4c06aed21a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-11 18:30:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6abdc47031457a31f95f0d327cfee4b1ae64a7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:03:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4d348a06b34a588c52cc838f6d9365\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/89b011351b4b4e6e85503990ddaaf508 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dd6ea262bf1488a8d905d3bb6f260bc : SUCCESS in 20m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b766908822d1426cb71693f8e7475c1e : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0eb9c0e925fd924369d32da7561c182e4197f83","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-11 19:13:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc8d906311bf3d3e221450d111f9cf3418ba3a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4e8235c9eb4b00b5cd76c0ef82c0ce\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7cd8d74f99b14f0f906f79908f692e2c : SUCCESS in 1h 06m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/08db3a00924f4d71bf77c2d46f376a75 : SUCCESS in 1h 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfaaed8052804f52b99c41d7451b2589 : SUCCESS in 41m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaefde826add4b8d9b2afe46af568e58 : SUCCESS in 1h 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c614cb493d943f3b05c4a60417a098a : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fe63cef94fb40f5a991c471ebc31556 : SUCCESS in 14m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/331e4d8e23f6492a8856ba6d20b20b3d : SUCCESS in 11m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ceb92a6b5ef4f7d872969d3f7edf4c7 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09306c398b0d425c897bc7b05bfbcdee : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8263faadc5304426bc58143659c47f9c : SUCCESS in 8m 20s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9ebf481873144a28a4affaecf9c414e9 : SUCCESS in 1h 09m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/734e9d13c0714cd1ba175a3383e7883c : SUCCESS in 1h 44m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e2cc0133d024750a71aee1eb042e268 : SUCCESS in 26m 24s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/449d66cf49cb4601a7a68d76f50951ca : SUCCESS in 39m 05s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f61e261e66584c31a2845fac25222daf : SUCCESS in 45m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/43aedc3cd1fe43cba65fab65e38e200b : SUCCESS in 26m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e484ca60e72c4328b7c5dd237d4a4b9e : SUCCESS in 1h 44m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/3cc4d6fa142445c8b2718aab1e0dc812 : SUCCESS in 1h 54m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8c18826fc61e47a7aeffd922648cd968 : SUCCESS in 51m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1b222184b9774dfd9357df0d816d6d10 : SUCCESS in 5m 08s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/35eb651556be4f18b2cbdeea41c70de1 : SUCCESS in 28m 36s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c9cb4bb59f5f4b2db36510c2f3c8f88f : SUCCESS in 27m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b9355dc0897d45f5a81654b3b28edc78 : SUCCESS in 30m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/04abfd8b7f4f43bc82403ea098e60f6c : SUCCESS in 1h 39m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a622fe8cbcc4fa1abfb9dd0d23960ab : SUCCESS in 57m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe574f6f6bae4246bec41433e9ed97a4 : SUCCESS in 52m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8d7872af697d4f329c169e87174518f1 : SUCCESS in 35m 19s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ef273f7969954c10bde9e2924a1cebf8 : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"440563194315cd56894fe7950f6afeef18f52c7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:33:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbe9db8ebd196ae5bee131a59d53a8840787ce3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-11 20:41:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"92491c5a9999a9f9e712ae7d3e2b5dc2311c7e1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:34:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea8859665f544f2a84f6004c8c9d3070\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac8398e98a5f45cb953c977d7f7dcc24 : SUCCESS in 51m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/65a925432f4b4ec5a34e6714b18ffbb8 : SUCCESS in 1h 54m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc40254762f642168307102d7d0152a9 : SUCCESS in 1h 01m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244b2e940c0a4967839be6557c218e84 : SUCCESS in 9m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d503b79d1d74b41894056c0e15c02a5 : SUCCESS in 13m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0079a1f30cd5422ab4d716af79f7ae80 : SUCCESS in 15m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084e01ae8cf445af89a994c224947693 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0803d51d860f476fa06e3f0d2143f9a0 : SUCCESS in 8m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8704965809f64eeaa8c1bc0c43f3cdff : SUCCESS in 49m 24s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/03b7734f75d5435eaed395bcf643bcd4 : SUCCESS in 30m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d5ca7b75bbc24da2a32f1ac279179acc : SUCCESS in 25m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3d393ac4d281491e949dc01b0c81f46e : SUCCESS in 31m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/137a680a0a5e417b92d11ead2507e999 : FAILURE in 1h 55m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/37fe2fb5c30e41f1b22f3c563e4313e7 : SUCCESS in 1h 37m 36s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b4da91be29d14a6ea897659b640fb15b : SUCCESS in 5m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5865613510eb48d1b9e59cd808ff1d53 : SUCCESS in 1h 42m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae5ac651bfb74bcaa75376453d3cbf6c : SUCCESS in 1h 00m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e20e37bb7e84229a19c36f775b50c7e : SUCCESS in 1h 01m 30s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/11926b258cfb440aa8a21ffd1760d2a7 : SUCCESS in 12m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/af88559ef21e40d6a1a401e1e9e180b4 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"be740a21d06c0e30a80d147ce7b444b2702dbcfd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 06:55:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93a768eb4535326e2b22dceccee681ab5be4fca3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 07:27:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c49834b6346843c1ae6ebfa68219a1ff\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a29fbb8946f4568b9a5ec334134f30d : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5cd683346af640e2b3543e73fa0833b7 : SUCCESS in 19m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eb4891fd2dab4ab79d911f711e7fb747 : SUCCESS in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38cab79848d86eb0cf95fb9ab3c3faf8a3bd258e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:14:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4f76b60bd994fc3bd991b54826fe9df\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68ef1492866946d6bb344d17486cdbde : SUCCESS in 50m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/72dafc32948f422281176bdd9d9d69e6 : SUCCESS in 1h 14m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff0e7c763d9f4a81834932b3feaa70e1 : SUCCESS in 45m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c7c4e20f7064fe8967106ea855d29cc : TIMED_OUT in 1h 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab0c4db8b7349b68a1e1397f5e65707 : SUCCESS in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2e4577531c84a829da627131dd59f01 : SUCCESS in 14m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61e04122ffdc42caad46365da2b56421 : SUCCESS in 11m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/55fddef2a7ca43bd8a19126ff74b3e2a : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23744a16e501464d8e64ee67d7611c7e : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a69399c2e3854997a07109d247091092 : SUCCESS in 8m 17s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5ca7da802e6140348d9cbd0c6b042863 : SUCCESS in 1h 07m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09f8b31df094430f92dd8c3287f96f33 : SUCCESS in 1h 49m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/32cc9afd334040f2a3de771a5a842927 : SUCCESS in 39m 37s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6f344e1de6e7442192bfd9cdf099c507 : SUCCESS in 39m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2db4ea67a0bb466ea02f6a690c925d93 : SUCCESS in 57m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ffa51876263a4447b5dc3f4b69d34f29 : SUCCESS in 51m 17s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/413f629dd9364eeca59f2bafa1fb2cd2 : SUCCESS in 1h 59m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/fff501d7f58e4b7693b9ff6d8839b16f : SUCCESS in 2h 11m 38s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c6ad0a1636404a558a26b5401ab02831 : SUCCESS in 47m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ba30a9e4e4544b0294df73bd6b539833 : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/08bd2d289f1945e9849e66df8ea65e20 : SUCCESS in 22m 02s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d3cb39083a39465681f5b567ac0df9a2 : SUCCESS in 23m 17s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0495086a061845498a2070c4ac86fe23 : SUCCESS in 33m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/06e29dd75ad141f8b3027b7280606115 : SUCCESS in 1h 40m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/073470f506b548e2b9afbf57cc32eafc : SUCCESS in 43m 28s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/cced2cc72d8e4f06acaa1d5d7f835683 : SUCCESS in 56m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/041ab2b439314006986b7d5fdb9311c0 : SUCCESS in 34m 40s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/259562c6c025421fa89812c3c4a6456e : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e6d9e4704434bc0c746625039006aac7e6c261f6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 09:30:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46c48be428e0e2179d814b7b33d8ad16ac746f62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:50:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c20b8583bb4f7884f1b904275fbe8d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a0e31a0bceb4eb6a03ab41dc4948710 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bde519f9dae3427d9bfe50e511336146 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b9a9c7b19fc3477db80b4ec90c8e6d3f : SUCCESS in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2010cb9bd6d6b4ce0f46e6a7f69aa723ac32f69e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 11:40:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b0b49a3717441948150e6f5515f4664\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/955c6e17d6cd4d95982f8238a086b021 : SUCCESS in 34m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/61408863881048969d78b0f8b5eb2bb8 : SUCCESS in 2h 03m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0eb27137f404bf1b11481c5a0d7cb4d : SUCCESS in 1h 02m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d005b15a659c4c8880b4ea10bd211bc2 : TIMED_OUT in 1h 41m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b2231952ab45ebada4ac3784304c3e : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60c156af6b0c452aa45e55b992ad8bb1 : SUCCESS in 14m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b42ae7c2422d4cca8e0bfd523075a40c : SUCCESS in 13m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12809c83d86a46ae914b63e4b0fa708d : SUCCESS in 17m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1284747732ad4a198d935e07b246dbe7 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab24948d781a4563a32a8e39b600baca : SUCCESS in 10m 08s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7104bcd62f7d41da9e6783125dc368bb : SUCCESS in 1h 01m 37s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d15be57396d14b249ec71ba92256129f : SUCCESS in 1h 26m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c61291a17b05402bbded17e3f4c08aaa : SUCCESS in 49m 27s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e6748a5cc07c46d3a4601c0b4192e28c : SUCCESS in 37m 01s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a425b1dd78da4c349d3a12b342bb721c : SUCCESS in 1h 00m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0af8745184b3482bba1bdf00cbd1a97a : SUCCESS in 53m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/55d3b7b449ce4855a6e3595874877acc : SUCCESS in 1h 34m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/fe2dc9fb88044cde809e8a07e50a493c : SUCCESS in 1h 57m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/06edf6c71abe44b5a239b875a217b7da : SUCCESS in 46m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/87a8911b28f04b7d89d18aa89d014360 : SUCCESS in 3m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8279452514ca4aa2b690c016f7b40dba : SUCCESS in 28m 25s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7ba7bae1ff3e43d19eec2663770f8ea7 : SUCCESS in 27m 07s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/001f2b08e868476aa3d0c68062ddf53a : SUCCESS in 30m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/897ab2fd630d40aeb42845ba9304061e : SUCCESS in 1h 25m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5394e233ad247b599da026f01827840 : SUCCESS in 41m 21s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a1bdc9fdee9e4629949ab9367e7508d0 : SUCCESS in 54m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3fdf781d4ae74caeb9713bcf2e7ebaab : SUCCESS in 21m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/68552ece214e447fa1d80b573c9764ba : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0822c18a00723e89a1fabce4f12b8597bb217356","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 13:17:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c956e3db11b9764b83e02615a28cb2475adff44e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 13:18:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65b0151299b6ae458295c517e6c816234e90f42f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:44:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2852a26ae48d492f898929a6386c3341\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe311073c37f4dfc96ab69f079b0dad0 : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c317547de964643a6559c16e840b72d : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b2b1f21d989d4281a8b1d140dc2c4cea : SUCCESS in 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fa5a31730ed23bcec7f0a44ae50ce6c3d832381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 15:25:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/163376babd4c496db60fc88b53e1dbcd\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98b8164e4f984921880caea5492d2949 : SUCCESS in 1h 02m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/af9376b364604c96b7d28790c92612b5 : SUCCESS in 1h 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd1d9ac3d37e4b8fb53e2f7700a87c3c : SUCCESS in 56m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee1b1292bcde41aa87190c774466292e : TIMED_OUT in 1h 41m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62dd4a1ba1a4a9f863a4b23d9f84e6b : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53db8dc1585c4e71a43996aebe057751 : SUCCESS in 13m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dcb18c0fc2144c0a1938fd6e20811b3 : SUCCESS in 10m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81fd510d8cc74ba69dccfa6267d17c4e : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c4a118b40c40bd97e39cdba35408ad : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e06bb3d82f8c4dc58916dbbae88f66ff : SUCCESS in 8m 22s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/14c0f41a95be4ca28bc6a380f24c2268 : SUCCESS in 30m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a2d8e3f70624d3bb357b2a53a052438 : SUCCESS in 48m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/799e912bb299430ca8aea0cae3de26f4 : SUCCESS in 40m 11s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f8318f53a1a141658dbb7e921995d085 : SUCCESS in 38m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b510800debd547cb8db66c3bba555247 : SUCCESS in 38m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/02300d22d9c44f75b1f385d80eb7ae98 : SUCCESS in 48m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2f3789d5f7fa4dd9b83ed46b0e95d705 : SUCCESS in 1h 55m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/a11b0a1231e3481eb9d343b615ebbb98 : SUCCESS in 1h 57m 21s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a7177c2f002f4fe8aed3d47ebdd129cf : SUCCESS in 50m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b7e2cb870a68412d829c8064c0f8a6f5 : SUCCESS in 6m 09s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/73e5bf44d4b648e09dc08e3993a57edc : SUCCESS in 28m 03s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9a3cf2b9f70c4fe6896403d5b14d4e84 : SUCCESS in 29m 51s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a2dc1c293014438a90c751f26e442e1d : TIMED_OUT in 2h 01m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1d6b3f0ecfe4ea9aa631277bf0ce156 : SUCCESS in 1h 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/041a6bf7410643b2aa22277889dff5bc : SUCCESS in 28m 02s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/89b931a46d19407eb4bee5d2fd824ad3 : SUCCESS in 26m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/06b6ede097584464bdf4da4be7b9f646 : SUCCESS in 19m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6cf419fe502f4c46b380b69197a0507f : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e7a50b3eadde2ca8083f27e935d394230ba6b21e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 16:02:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9200089b07899a274884568728618290aa74ee86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 16:38:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18dbecc86e204480b0c5a13fc938fdbe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01e1aef6f0314d48827a278678f78888 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6c36828af5944e799b70382d062024eb : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/14720051046d4a8d90439c019a617093 : SUCCESS in 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85b4cf326c7321f5b24d9f25caad4f2ca12e2e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 18:04:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ba9b084bbd497cb78ce328314d7d9e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b0905046a624d89ac7eec02fb9dd7ba : SUCCESS in 1h 02m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3f2ca101a68f461daf0492ce69b0ffa0 : SUCCESS in 1h 31m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76dbc2c67d944aae8f65180aae9f178a : SUCCESS in 1h 03m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e102089ae58f4c7fabc7b3da321e46c8 : TIMED_OUT in 1h 41m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8119e254cfa349129848afcfdd34c0f6 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ceff0153eae460795260065ae4950eb : SUCCESS in 10m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43fb9a3900d24504afe03253a314abcd : SUCCESS in 13m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/482f801c825d4284a9b510b78097520d : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5eeb4c8cd24c0887da903e71c3f093 : SUCCESS in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6160c1cd0f34cd9b43733dc950205c5 : SUCCESS in 9m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/81ee0af34d394762a292c134bd8d4d22 : SUCCESS in 1h 09m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/653ee8f1839d4f3082add147a6b31d20 : SUCCESS in 1h 28m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/380d79cc02494819a7c6147212d2c845 : SUCCESS in 48m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/2056a6e31ba84c81bf7658819cb2612b : SUCCESS in 38m 30s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ca27d82b82c84d43a1300f0325082ce8 : SUCCESS in 1h 00m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/88edcde994de41c689a3d84744ca4671 : SUCCESS in 40m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/40f3261272984acba4f468f9618f38c5 : SUCCESS in 53m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/1b75b5348b3744a992b7ca5588e457d3 : SUCCESS in 1h 55m 29s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/371917826409483f83fa6504e0941101 : SUCCESS in 30m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bbac99f4c90f45d98712e38250056951 : SUCCESS in 4m 27s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/56541b5eb6204061b4fee9092a3c0b57 : SUCCESS in 27m 54s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b77459674b54495f8f205fd1a941b9f1 : SUCCESS in 21m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/095e10898b634ed2ab57483f34882fa4 : SUCCESS in 34m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb42c494c8e348d98a6a64904bf422fe : SUCCESS in 1h 05m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54158c8a9a3144488d8e6d9e5a758c67 : SUCCESS in 38m 45s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/045cac6770ae46fb8e85697f9a5188d1 : SUCCESS in 54m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e28eb80dbd5040c8822f3ac405bfbfdc : SUCCESS in 37m 18s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b928dc08326742d8be0d3a5f20ceb912 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d047dbad0983f5557d3b61d866d03c24377d8e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 18:04:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31b6db4a88a7f13fe103f84da8a0f6c0f010ec7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-12 18:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"aa86b097f28dc265bc188eb7b9f340dc7fdb8d45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 19:56:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53cccc82428641a28af597eda7af7bb9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b81ea03593544231aaa71010dc0b7f17 : SUCCESS in 1h 01m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ee2cba5923b84dfd97fe9d742378a987 : SUCCESS in 1h 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e4dc225a38843d98f61bf514be58971 : SUCCESS in 41m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e731d1783d49aead1c0f36d3ae183d : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77a45468eb06442c99b8c67dd4332e39 : SUCCESS in 8m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc7cae9343fb44638160f146ed97773c : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda5f775ad5b4dbb90d68a4c616c1b68 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88281d051ca14cf49c96f9a3424bda1d : SUCCESS in 8m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f1fcc6bb37344247afbe6d27226e9cac : SUCCESS in 48m 48s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a12e0e18ecf2458781fbef3c9fe8f64b : SUCCESS in 28m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a5e07bb691ae42528164e72866ce470e : SUCCESS in 26m 12s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/08bdb887ef224d159721c3332193121f : SUCCESS in 34m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/05f43dedc7494e94ac9f6891d1c85e4a : SUCCESS in 1h 16m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/87b9e3632035493e906e509a2e399d61 : SUCCESS in 1h 45m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bee5f285b71d473985125a3bb09f51fd : SUCCESS in 5m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5c7215eaf1a41aab1c134ee2d34dc3f : SUCCESS in 49m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3885688d64f4855821eecc937eb56a1 : SUCCESS in 55m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f76b843c41434fffa76e4d97cb05f0a9 : SUCCESS in 31m 43s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e6353e6894ae4480bd48a57f24f16f33 : SUCCESS in 9m 48s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9fc468a3e0b242eab7c27c5b3980e24c : SUCCESS in 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2c74c22895baa28c3f9a704a0614e8e54c316716","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 19:56:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd1c64d1c5650641a80eb66e842fa74d726bab99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 19:57:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272bb2acc7744dec8ecb8a94eeee21d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e297e568b732487caca230382534df53 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1976633a871accaed7372ef7c55555824326ba51","revisions":{"1976633a871accaed7372ef7c55555824326ba51":{"kind":"REWORK","_number":1,"created":"2026-09-07 14:24:14.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/30/1004430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/1004430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/1004430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/1004430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/1004430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/1004430/1"}}},"commit":{"parents":[{"commit":"1dc44b5988f75683564bc4e765be1f84c3d75dd4","subject":"Only load same-host instances when checking share usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dc44b5988f75683564bc4e765be1f84c3d75dd4"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-09-07 14:22:31.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-09-07 14:24:03.000000000","tz":120},"subject":"Add service version for Hibiscus","message":"Add service version for Hibiscus\n\nWe agreed by I2dd906f34118da02783bb7755e0d6c2a2b88eb5d  on the support\nenvelope.\nPre-RC1, we need to add a service version in the object.\nPost-RC1, depending on whether it\u0027s SLURP or not SLURP, we need to bump\nthe minimum version or not.\n\nThis patch only focuses on pre-RC1 stage.\nGiven Indri won\u0027t be skippable, we won\u0027t need a post-RC1 patch for updating the min\nthat will continue to support Gazpacho.\n\nChange-Id: Idd90094fd766b6d8f1bc6173600fef14e583a6f4\nSigned-off-by: René Ribaud \u003crene.ribaud@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1976633a871accaed7372ef7c55555824326ba51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1976633a871accaed7372ef7c55555824326ba51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
