)]}'
{"id":"openstack%2Fnova~1004658","triplet_id":"openstack%2Fnova~master~I13837d1b3cc09519ac95586553666b69854c200c","project":"openstack/nova","branch":"master","topic":"bug/2166786","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-24 16:21:40.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-09-21 15:27:35.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"hashtags":[],"change_id":"I13837d1b3cc09519ac95586553666b69854c200c","subject":"Reproduce bug 2166786","status":"MERGED","created":"2026-09-08 15:08:13.000000000","updated":"2026-09-24 16:24:06.000000000","submitted":"2026-09-24 16:21:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004658-bug/2166786","meta_rev_id":"0ab8978b5021d54deee8477db6b1a0a616e21e48","_number":1004658,"virtual_id_number":1004658,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 16:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2026-09-23 15:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-21 15:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2026-09-23 15:23:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 16:01:49.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-08 16:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 10:33:04.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-18 14:06:34.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-09-23 15:23:14.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"2eb18c7b06b71c2b41ac8d5906677bf460aa689f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 15:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1bd6361175973147d22054d74bf198bb903db50","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 15:24:32.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"befdb940efb39ac7d0d292f779dded4c18d7bd9f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-08 16:01:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":1},{"id":"2d18533713192ef704dc1a7646ea30ac6f3adbd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:03:40.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4660920c267448d91975fcc0cd1d1cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be651de267aa4e878ab66c567b08fe56 : SUCCESS in 54m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b2abed78044b1abef45809b60eab14 : FAILURE in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15895b0e26d1447f9afa0eb07ce20838 : SUCCESS in 15m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c43608232072416aa822032837d48485 : SUCCESS in 11m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14aead16e43842c7ad21fc6561d91b28 : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48be7318b3ac453ebecd0eff9c6ddcfd : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d751b700ce5c40fdbf673dd617455483 : SUCCESS in 8m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7cd590ea07724032bcbe5b89108e6965 : SUCCESS in 3m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cd0224d97b094f7f845858e5e072c554 : SUCCESS in 30m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e5d7f0f87674403e9dc8f88f20a6656f : SUCCESS in 27m 56s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c0c859af17d44988b90fe8d9c2b71379 : SUCCESS in 35m 45s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a6313090665c4c93a0092a5f861e5f90 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"727cb30c7311a716e891df32a090cb5a8647f80b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 16:07:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d776bc822a7b35ca6558f302ede259f9110142c7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 16:08:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f7c66c69de8e3d65a5124f0d1774a98adbce996","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-08 17:00:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":2},{"id":"95726efdd312888e85937fcf40bfa0039786fcf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:12:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b5e30fcd6e74be6ab66a2641d2bd1e1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5e9bc835deaf442f9f881743dfd60feb : SUCCESS in 24m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/220ded15409b47c88b2d27d7267989da : SUCCESS in 25m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3ddaeb4b319d4ee1b12789b4f829b169 : SUCCESS in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da83b55ac7674cf7c0738159e31d6705c01ea198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:43:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b16a450962564184a93689b47c4e173e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/787c654294d7471aac337e90f793f2f6 : TIMED_OUT in 1h 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a749dd4ca84fbd86085b75fb6f9540 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/587e34df152140db9e9eb4111d175291 : SUCCESS in 12m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6c60e8991d9419da29b653acbf580c6 : SUCCESS in 13m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c075aaafee84a45be7da1f01e6959ba : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f1fb6c35db4389ba90cd2d7c3712ba : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb8afc6b52034aaeaf3845b5959b3066 : SUCCESS in 8m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/461ad3f7c2ed4c21894bc580b0e3661c : SUCCESS in 5m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f0691ac2e5834cf895ebb282e976a6f5 : SUCCESS in 30m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/39b5c64da8264a5094b6afd32e63c4c3 : SUCCESS in 29m 56s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5cd57754c37447a7a6f3c9469e4cf211 : TIMED_OUT in 2h 00m 58s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b1beb3d329554df8a69dd5c708fce4c6 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6e8e6c6cf8b27ff63c97fbd94586afbd32ea34a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-09 12:32:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"35c13364b71e556d30e80b1cdaa0b2e935d6cafa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:59:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72efaafb30314cecbf2a2885de3d4efb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be97a52f7c5a49c09230bfcb8af7c5ed : SUCCESS in 23m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ea56c6707f042bd8a0c6e5f5bbe7db7 : SUCCESS in 16m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3e89b7ee7a0f4955943bcac179c581f5 : SUCCESS in 25m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b01c90ef25103313c2e4d62d485bc2ef532e590e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 13:27:33.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":3},{"id":"2cd34ead30a3a9d49118b67bd6c425ecf5ffe1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:48:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/349bea685a514b84bf309a0cd2503d58\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2138dca223541ef8193871b40efea2f : FAILURE in 1h 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a46dc09714e4037897f8373e923622f : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b30b2953a3843948deea17a80383597 : SUCCESS in 13m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae91ef06eba945b4beeeb0ec49b34783 : SUCCESS in 8m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e79d0bcdecb454d8e6104515c0acb9a : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b13a0e35ff45ee86350674549f18cc : SUCCESS in 14m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/411d8159c3624acb9778454fc621cc94 : SUCCESS in 8m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/138a02e7ec194f86b20b570ec8500624 : SUCCESS in 6m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b165002d97242d0a522d24128b37bba : SUCCESS in 26m 02s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b354e5866deb467fb3352158fe54c23f : SUCCESS in 27m 49s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1124ee77c941432fbd7058e65f736cdb : SUCCESS in 31m 50s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2ea60439f64c4768ace93ddc5754c162 : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":3},{"id":"067118e5449fab19129de08d1c7a0745804f7859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-17 09:27:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f45341a5d17780a7d5744cec2c7e48c092cb103c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:12:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7d1d444ac84d4e9c208568dcdab1c8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c5003cf756143d69731eff98b918ca3 : SUCCESS in 21m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd5284c219ec497a9bb8928c15f0ab1f : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eeba29cb1c96452f816ec1a6cc8f4656 : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f42d1b5d5b4da374cd6f131ec95839e8abfd56c8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 10:21:40.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":4},{"id":"aea8ddf8236cef312011bf366be7f18e8f1be3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:32:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c833bc3bec6e4fb58b519d177d81d9f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26acfdf7136f4dd997fcd05411158d87 : FAILURE in 1h 01m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c1cb9360084a279c5379f6c5b61229 : SUCCESS in 8m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff475fe47c804c8497b22c99d36d6f29 : SUCCESS in 15m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e87464a5a684062b59789fe5cdd1931 : SUCCESS in 15m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec54ca9844734e1180e7e0536c341123 : SUCCESS in 19m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ff510e679842a1bb3154f824ecfe41 : SUCCESS in 14m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e5c4f7a591412c93c420eff67c7691 : SUCCESS in 10m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/74f610e827f5404c8b16c63a9fe155fc : SUCCESS in 5m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bde061dce9e547609053b200efba7909 : SUCCESS in 26m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5743a029a8d941ce9cd1950e122e640e : SUCCESS in 28m 11s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/bbd3f75e2e484e1f957cc91656258af9 : SUCCESS in 37m 00s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3046dbc75ee74413b29e7f858b1f7aca : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":4},{"id":"82b962b15a189a5b35caf945f6c4bff4669b9759","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-17 10:42:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"600d18a6bfeb82736e38d2c9d054d885ac24a6a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-18 08:30:08.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"1e01b3fdef666559b8dc690218e7353db8f51731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-18 08:59:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"42dd8c1fd4652fbc4618236a1a2910265abb83f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 09:23:47.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":5},{"id":"d7eb8735f3ec30ed5547d72863c9044471a8d36e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:36:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de3131ba043d41c699669b189891a5ed\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/16abdba8b4dc429a851bb0e82dde8df6 : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9a24ec643c6149d68b64bc95924d0ae5 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d5bac1d2e9d3473c8bad7ae090bb0dee : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5bcc1a119c6c16aa1d50262b464e4e570c1f8759","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 09:56:34.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":6},{"id":"ac792bf30a79e5d5f367aa05b41a53e9497a59bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:33:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e7c7f9a2a78499e9a79a2b61ff5831c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2abd6ccdddce40799ddc05ce6bfa7b3d : SUCCESS in 1h 25m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b0dee5ae1c4eeabd1f6b00d80d7835 : SUCCESS in 9m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/940a131945114995866f5360120d4949 : SUCCESS in 13m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7872c7b34e8543f38dc81e138a6b9b32 : SUCCESS in 14m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f81f3197fa24df791d723a878c5ccbb : SUCCESS in 25m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823348a812a54670a9dc575acf5f3736 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6385b52723304890bf9785aaca68c678 : SUCCESS in 10m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3abcf3947d3b476399e789ae0d0cdc43 : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1318ff249ab243c9ab7c3cad987236f9 : SUCCESS in 28m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1a0194f3dd8b4e7990aa9ce05503127c : SUCCESS in 29m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b9d63439157a489496f3376a33b25272 : SUCCESS in 40m 10s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/17821bdbbb164f0c89d9a0e07635354a : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":6},{"id":"2caecd67a8b6c9a4ac7a1f6dfd7e4b9c527291ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-18 10:42:55.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"508e25d7648b469bca5240fbcf5825b680365c0e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-09-18 14:06:34.000000000","message":"Patch Set 6: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"648517664acc1b5889908c244ca78e3e086d28d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-21 09:19:01.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"d946ba24a2e6d6439e1367c50ed50c4815fb4063","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 10:14:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570113d85dd244a4a862af3a9c759c34\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47363dbf845f491cbdaa36844a6096ad : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/342ce0311f544927b35f9d63ba954851 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b9254d351f89489fb7a71961fce57c5f : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8fa3b4a44eeda43db2e05cf908ca5810af988ba3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-21 10:15:22.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":7},{"id":"d35b304dc7139a3032d96da3d9c0e7215b243306","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-21 11:05:45.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dc784c6d0a929e3b8fe1cde058224b3e5e75df4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:47:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/059c8ecf39af4843bc8776e1501936fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94c0f2fda07242d4bb837e66ba37b456 : SUCCESS in 1h 00m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38aa32e26504a96bd359e1840b72fe4 : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72c4abc5e47d48d895e6921da5a1d610 : SUCCESS in 14m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48dcdc5fbac74447b9ce9fbb3f84bd24 : SUCCESS in 18m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8dd653aecdf54ac98d7026a18b3ccff9 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae983a39e82c4854a1a92124f92da88a : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06dfa54181944f7798774c2ebe590a0e : SUCCESS in 9m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5c2aa98d2e264a53964eef321d93bf22 : SUCCESS in 6m 01s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4d28065a9b5e42b5863efddba91b36a3 : SUCCESS in 29m 29s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/359ad8c745754a36ab87b9e92d8a58eb : SUCCESS in 30m 56s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8b73435136974f26888cf82dc56c3efc : TIMED_OUT in 2h 01m 20s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6364108d0551462dac77e9fa80bac87b : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":7},{"id":"356c3647ec7dfc9ba9709c6a366fbba937cc6410","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-21 14:05:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"df4c0189265766d0ad6597afa2413870ea39cb0c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-09-21 15:27:35.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7576b2aa4114db3b5791b3051101e533dbd7566f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:29:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ff98ce68b54dd0b60be45782b17f9f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4dab7931e15d48688bdab3b05e128981 : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65f34e9b743c46e1bbba28c924b6a3de : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c0bd2e2962224b3794af61aa0121aeaa : SUCCESS in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e840a3425edd4c6c3aaecc754cef3cc0d0919f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 16:18:37.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e3691de4a554b528abc4daabb95d408\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d5179059e5b4d85b6f4b4a0c7051e04 : TIMED_OUT in 1h 42m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0abea53818b41d48c3fd2d7591f21d8 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b18ef80c8cf6414eb8662a981332ddee : SUCCESS in 14m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b29840afe09427c87799ea9c319c48a : SUCCESS in 11m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d574c9df116a4dc7a4fc12adc86294ff : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91cde90900324375a22a277946dfad40 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e90db55cd2114ab794fd7965db66f0c4 : SUCCESS in 11m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1a98d242c1c64e7d8b575f0ba46e59d4 : SUCCESS in 5m 32s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b69b40d7be1f4096b8168bad6b8c9214 : SUCCESS in 31m 33s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/89f5379c03dc45ba86f00255343263da : TIMED_OUT in 1h 01m 42s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b6539a5e89aa4e65bb1ee8095ad630aa : TIMED_OUT in 2h 01m 40s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/df30d731cb7648888e50bf067babcf6f : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":7},{"id":"3985d3eabc71e3a2f6419dd3f19a5e5412fd84fe","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-22 09:18:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"629254f9ee3cdd4d25ab8b9aaae6c6c4ef1df82b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:53:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3578d03af5c2410bb56e2bab3eecfe64\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5e37866b49314686a1ec2eca86bde2e2 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad26fe1b08674d73b2268b200c0125d6 : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6dbcede91d94432e8cf6bb86889d8687 : SUCCESS in 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d92c23a22f486b31ebaa204832aabbd4df23f7ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 11:31:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9274977842ed4d4491cc7ba936c1c5f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86a2577359e74b50a95e0cf9da224a4a : TIMED_OUT in 1h 41m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/845244e45e474d36b1fafcd85f7e5db2 : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cec19583329c40028b2f79fdfb0aeefa : SUCCESS in 15m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cdbb641d8464fdda18a4ac0ea172504 : SUCCESS in 10m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ebc3013a61742288f1bab163c6bde5f : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3c426a14484cb2b83b3d0a3acd8a42 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba007b4111c34e04bd04c4256e9c1271 : SUCCESS in 10m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/298cb11ccffa4018a763eaa24619df68 : SUCCESS in 3m 15s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/74735edd4d4e4a2b9db02c01e5c895ef : TIMED_OUT in 1h 01m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2acb727cd7c424d94e6ea9b7f1f0fc7 : SUCCESS in 30m 27s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/29390c349ab94fbb9b1441bb40b75b26 : TIMED_OUT in 2h 01m 27s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b6a08cdc634c4f38aacb9fe29f71104e : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":7},{"id":"b3ad5ca7e2009bbe011ea8261fc4a233a38e1775","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-22 11:57:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a75abd176d00736be45d0d0ba2e806fa1f131b2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 12:42:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90df41ff5b794a1e95e1f10d06dfeb33\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e111fb0d0edf45409bb17b2383220f6d : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13e9e99321884a8288c926ddb39399bc : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f7efc6e5288f46dd9d20ad05f7d2a7ac : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fcd09e8becc2d71b02a5a66940bdbea533288cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 13:04:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba6f0df51f346fcb7e0c2a9038afb3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8770b0ac159b45928a14eed3cbcd8da5 : SUCCESS in 54m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f76702bf7464ed8a3dae4c9f0702d03 : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5db37c9ab8e14e2a8f9a57843edebff0 : SUCCESS in 12m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/812744523350475bbbdb83d598bc6495 : SUCCESS in 8m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/301d41222f63437f9ac6f51ca8972dff : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df013eb2da67458f91ea321a7e333ad1 : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/093762237e884686a2161ec1698033f9 : SUCCESS in 9m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9633e961609140d0a6cf426cb3c9af88 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9db9faa30a6447209ef02094cc90f3a6 : SUCCESS in 36m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/087f0d30f52b42e8855eac119038a2eb : SUCCESS in 27m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0ac686076b8b4a6ba409baf775a07715 : SUCCESS in 35m 36s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c137a3c3ce1042a3ae7a531b3ef1e9d0 : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":7},{"id":"05fcfe91ae3f96faa152952aa9cb9633c606686b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-22 13:14:26.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"d7aebe187aad1cf61fa31ed993bafe28a5c24199","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-09-23 15:23:14.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8e7df09cfddf1ffc6579afa9c19afce1bb83c227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 15:23:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"01fea5f31ca8024bb0376c60c8f567b4a4a9178a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 17:29:42.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c0ab5ed3ae3468395691651e157f810\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089291badaad4d8298e3dd5f1f021f7d : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e68033bd1c045a79e5807bebb187a25 : SUCCESS in 12m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52c12daa9fc34a33ba51d988a6f1cd34 : SUCCESS in 15m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b558ef940115497cb80a88e64ab9b3ec : SUCCESS in 14m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c037fbf3acab41f0b4b267a0c2818451 : SUCCESS in 9m 39s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/20ecfc2d819a431287057923dbae56a3 : SUCCESS in 27m 41s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/46ca46c0eacd4e849e9374500d47367a : SUCCESS in 27m 12s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6b3c1f5452bd4ede816fae1c46138d9c : TIMED_OUT in 2h 01m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3583bb10092c4c028b3c6c5442cf6b38 : SUCCESS in 3m 36s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7f8b8a26f918491d8e1d39788f1a1ce8 : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":7},{"id":"56fbce02b608d06719e7801436c2934ee940e6b9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-24 11:15:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cfa7f18d09adfc87be759019eb26de59eb905030","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 11:37:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19a7744b88c4cf4a870af916ffc3d44\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/337fae858d3240eabcfe09173e1363a5 : SUCCESS in 15m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/15cec1d0c5fa49c2aa36d4c3f4223be1 : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6c32097bcf9d520caa96e258e8eb26d95b95dd10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 13:03:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788db1bea9d942bd8c1035fcb80758bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c578d3267641d4bb6f7c1f6193609e : TIMED_OUT in 1h 41m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9537453b9d634b428c387d35154c6abf : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c34cef3b52440f5ba44870c56ff54c3 : SUCCESS in 15m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8fd5615f4d654264ab3eb012fe39da20 : SUCCESS in 19m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a209bd047157403fbf6e1be77d4aef1a : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d9142bab954420193365b29c2f14fda : SUCCESS in 11m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bcd25338db38416e9185314b9dfaa55c : SUCCESS in 5m 10s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/70a83edc94324f4b8f3e68702fba70f4 : SUCCESS in 37m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/08cdea32e2984277ac51911223fb42d4 : SUCCESS in 27m 56s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/efd597dd44074bdbbcfe4dbf1f87e67d : SUCCESS in 43m 13s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d8f82c43d78244c88cbc41fd85e44008 : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":7},{"id":"1a62df1e702913b4370570032ed9c0badc0f9681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 13:06:37.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f7297eab4659ef921400993358212225413ddc2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-24 13:13:51.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"81f13e4d294a1e420ba39ed979dc9ba255bf26a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 16:21:40.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c527c846fc0429892e5096422e1e9c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5cfc7abbfb4d07a03e66635081a5c3 : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a16ff86e79647fba404d1fa6ae8466d : SUCCESS in 11m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96cbd96575a8495bb5e7b59ff18e2e90 : SUCCESS in 18m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f68b216d092c49b29ad3bed38bd56624 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aa88ffd8f3c46cfae6daa83aed08632 : SUCCESS in 11m 01s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/437dd06cc5bb465fa68aa9298f8e69d2 : SUCCESS in 28m 36s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3da8eb12760a42a2b89c991cbba20aae : SUCCESS in 31m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/65d34cdfb24f4685935d0772c1c20fa4 : SUCCESS in 34m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9e560a7976284f948e05c5513b575464 : SUCCESS in 5m 50s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ef274465b95942fb82afa6cc47e877ac : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":7},{"id":"ddd0c18a4975466b429e4bc64e88fc23e74b391d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 16:21:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"0ab8978b5021d54deee8477db6b1a0a616e21e48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 16:24:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e4e9a80d49045c698e8159a5348579f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477e2a2304374c85a3db8e1e26d55dab : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fa7cf515b1c6c608617a887000192948e314f9a4","revisions":{"d0734154a8c405a520a79ef9d997581903f314fc":{"kind":"REWORK","_number":1,"created":"2026-09-08 15:08:13.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/1"}}},"commit":{"parents":[{"commit":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","subject":"Merge \"Parallelize per-node resource updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7582db6d65b9880e1ef7d906ddad1d7c95bb53a5"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:07:49.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0734154a8c405a520a79ef9d997581903f314fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0734154a8c405a520a79ef9d997581903f314fc"}]},"branch":"refs/heads/master"},"68fdbf253215edc20626fb9e44f7ad516bbf99bb":{"kind":"REWORK","_number":2,"created":"2026-09-08 16:07:35.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/2"}}},"commit":{"parents":[{"commit":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","subject":"Merge \"Parallelize per-node resource updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7582db6d65b9880e1ef7d906ddad1d7c95bb53a5"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 16:07:28.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68fdbf253215edc20626fb9e44f7ad516bbf99bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68fdbf253215edc20626fb9e44f7ad516bbf99bb"}]},"branch":"refs/heads/master"},"6de0636d87d7ecd3b15445746396ff0a505f8e9a":{"kind":"REWORK","_number":3,"created":"2026-09-09 12:32:09.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/3"}}},"commit":{"parents":[{"commit":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","subject":"Merge \"Parallelize per-node resource updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7582db6d65b9880e1ef7d906ddad1d7c95bb53a5"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-09 12:31:45.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6de0636d87d7ecd3b15445746396ff0a505f8e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6de0636d87d7ecd3b15445746396ff0a505f8e9a"}]},"branch":"refs/heads/master"},"2e6f5e9b22e5977abb72b64726ef374a86921174":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-17 09:27:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/4"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-17 09:27:43.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e6f5e9b22e5977abb72b64726ef374a86921174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e6f5e9b22e5977abb72b64726ef374a86921174"}]},"branch":"refs/heads/master"},"0c17ef4c0545e8d8bf4b0e276ccf63879aa5c994":{"kind":"REWORK","_number":5,"created":"2026-09-18 08:30:08.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/5"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-18 08:29:00.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c17ef4c0545e8d8bf4b0e276ccf63879aa5c994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c17ef4c0545e8d8bf4b0e276ccf63879aa5c994"}]},"branch":"refs/heads/master"},"ac3f4c04ad635f6641bdc45ccf21ac6fe6b5cc6f":{"kind":"REWORK","_number":6,"created":"2026-09-18 08:59:58.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/6"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-18 08:51:18.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac3f4c04ad635f6641bdc45ccf21ac6fe6b5cc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac3f4c04ad635f6641bdc45ccf21ac6fe6b5cc6f"}]},"branch":"refs/heads/master"},"fa7cf515b1c6c608617a887000192948e314f9a4":{"kind":"REWORK","_number":7,"created":"2026-09-21 09:19:01.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/58/1004658/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/1004658/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/1004658/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/1004658/7"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-21 08:52:46.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nNote that the PCI allocation in Placement (if enabled) are not affected\nby the bug.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa7cf515b1c6c608617a887000192948e314f9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa7cf515b1c6c608617a887000192948e314f9a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
