)]}'
{"id":"openstack%2Fnova~1004687","triplet_id":"openstack%2Fnova~master~I6f5bcf88684e50ab9941e93a63720b998f6a4e6b","project":"openstack/nova","branch":"master","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-09 14:52:28.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6f5bcf88684e50ab9941e93a63720b998f6a4e6b","subject":"move nova-next to ubuntu 26.04","status":"NEW","created":"2026-09-08 19:28:17.000000000","updated":"2026-09-09 15:01:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f235afcfb27206f2d016d06061044f4770fc0220","_number":1004687,"virtual_id_number":1004687,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 14:52:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 20:04:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-08 20:20:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-08 21:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 14:52:28.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"cdbc22a4d6a8cfd4d1db3e62f35ccc0e8250f57e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-08 19:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ff20b79947dc5fe3bf665084d9dacc364271711","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 20:04:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7eb4a7d49b042549b6b8c62f51ef2ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8386b83d2af148abb495669e6d412cb6 : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5196dd51f770490cb62c967de193107d : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/331e1d0b0334424bab6577d8eb36d094 : SUCCESS in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7f56e6c481da5572322b269dfdad6388322203e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-08 20:20:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6f5bcf88684e50ab9941e93a63720b998f6a4e6b .","accounts_in_message":[],"_revision_number":1},{"id":"cb83433ba1e3bbb6d4adc4d1094f97bc9ecce052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 21:39:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5ddb2e2839545bba2814da7c0d9561b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d46782b88d3406fb13a6fd304f7af26 : FAILURE in 1h 16m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1a401101b9c548ca9acc27923ae4de14 : SUCCESS in 1h 56m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00eb07b0fa7e417a8f719fc044da37ef : SUCCESS in 1h 01m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c0e451e18544149b66e64c3972245ec : TIMED_OUT in 1h 41m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47c5e1baba341a19fb63785d2d44f1b : SUCCESS in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e62823a3e6634bed8c9264af6a16121d : SUCCESS in 13m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8862544c2b9c4cb59192d46c05cde5ca : SUCCESS in 12m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a962f21249a43bb82cf010f2e941b27 : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95f6a46daf14cf9a77c6d5ad48618e2 : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaff6a51daa743cb894283a6064f0b31 : SUCCESS in 9m 26s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1ed1d2fa74834847b8307082d767f4a6 : SUCCESS in 1h 09m 13s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cbc7a77bca74523b550e33764b9c62f : SUCCESS in 1h 36m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/211dfd85bcdd44659d2fce109957e804 : SUCCESS in 49m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6e097ea18fbe4ed7bd37b3ba4522bcae : SUCCESS in 40m 38s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/83ddc124a1be4c4197a03055115d893a : SUCCESS in 1h 02m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/95c33ee9f4534aada0033b772ade8cec : SUCCESS in 52m 12s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6c0afe43e45041cdb34a88506d07b1d9 : SUCCESS in 1h 12m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/07f256b6922e4fbeb53d0fc45d58b3d3 : SUCCESS in 2h 03m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/1128b36b3b3d4cf78b38c9f57a1968fb : SUCCESS in 1h 36m 10s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/223a187a10ff407094d388adf3a44b9f : SUCCESS in 51m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/607752b92fb7417fae5d2d7fa06c38eb : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/626c1ed78114469b822383f5a9df180c : RETRY_LIMIT in 8m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c261b1265064023b48227db9e4d8d17 : SUCCESS in 26m 27s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0f00b45e481741bd839bab51f2a12aee : SUCCESS in 43m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ef9ab94068242348ee1da91bbbb3a3b : SUCCESS in 1h 41m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70006d4c7bd146c5ac69b85530937dbc : SUCCESS in 41m 56s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8f49967a74174950854a85c97704d832 : SUCCESS in 40m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0d6b82a5566542499c4abe835915a1b6 : SUCCESS in 22m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/229bb1a88a134203b707fbf2801f72b9 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cf413eb853aa0cc722c899555a144c98820e29c1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-09 12:39:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4011abb4af3f1230bf0f053b6fb03777c1a5b14a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:17:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfa72149787b40458774be68e1da6f7d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73a89019a6ed485f9bd396ea2362df08 : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e017024581774298a0d5b6b72fcf27d3 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/84ccf691caaa48c1b38d966140a2e76f : SUCCESS in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dca3f201886a74ff515f47a7090a1cfbd5ba5656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:52:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/746640b855a2497287181ab46ca497af\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0c26372185b4ac2a1cf4a88e5235452 : SUCCESS in 47m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e085e0f3bc1642eca0e17f00783d2567 : SUCCESS in 1h 54m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f2a1e47bcc044e19200479b2f30e923 : SUCCESS in 50m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d5d7138ad0847e397787253375dbc37 : SUCCESS in 1h 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c75e6874b7a4a749c323564707929d0 : SUCCESS in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/044e711b9e844e8ca0ded47e6278b42a : SUCCESS in 12m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52b493c11582433a93427fd44be02cd8 : SUCCESS in 13m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c7d585c9aba4672a3e1b581fb4bbb1f : SUCCESS in 20m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4844cf4f762740b78da2dc33cd88cfc0 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a6c3945dd14fcaaaa4085dd5de5600 : SUCCESS in 8m 29s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/86f257705ce548b7a2f2afe339ca8c0b : SUCCESS in 50m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2f650c4131854c6c8d49d208241a1f1d : SUCCESS in 1h 28m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ef681f4b35494846b75ca7dc613cf7ac : SUCCESS in 55m 54s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/54ae799f868c4ee183a9a7533a12774c : SUCCESS in 38m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/771400e5c4a2411391c6e52daf92b6b8 : SUCCESS in 59m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/73ffd88feec24673a12248c846b98136 : SUCCESS in 53m 52s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2e4aeb6d496844a899c15702478aa5fa : SUCCESS in 59m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/24d8946a2e40427cb9b70cd39ffebd01 : SUCCESS in 2h 05m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/40a759f658c148af8cb80aad3c5d3b6d : SUCCESS in 53m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a872a3f42de743f38e11cc4eed62f95e : SUCCESS in 48m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/03b0fdc663f6434c92551fd7ecf3f380 : SUCCESS in 4m 30s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cc004149f4e448ab8516182e05e26dd4 : SUCCESS in 25m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/71ce175f5aa9407e8b1a427d019f2c66 : SUCCESS in 25m 56s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3f195eb90d52486e8922b769e5fd7e35 : SUCCESS in 41m 43s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9756d58a75a449338bee0ccd24a3c446 : SUCCESS in 1h 32m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cf75182922d4a95b07bcaaecbeba091 : SUCCESS in 53m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ad7d77ca95b4a8182070432d4104691 : SUCCESS in 54m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5cee5d136b9843aa9dda0d06c2b1d5c6 : SUCCESS in 36m 07s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/af1ade137a1d42a3a6a937bea5cc244c : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f235afcfb27206f2d016d06061044f4770fc0220","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-09 15:01:53.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a1a05dcf4acd0749a3db64a4ebe50c4134c0742","revisions":{"2a1a05dcf4acd0749a3db64a4ebe50c4134c0742":{"kind":"REWORK","_number":1,"created":"2026-09-08 19:28:17.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/87/1004687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/1004687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1004687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1004687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1004687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/1004687/1"}}},"commit":{"parents":[{"commit":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","subject":"Merge \"Parallelize per-node resource updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7582db6d65b9880e1ef7d906ddad1d7c95bb53a5"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-09-08 19:15:21.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-09-08 19:15:21.000000000","tz":60},"subject":"move nova-next to ubuntu 26.04","message":"move nova-next to ubuntu 26.04\n\nubuntu 26.04 provdies python 3.14 by default\nand will be a required runtime in 2027.1\n\nThis change moves nova-next to 26.04 early to provide\nadvanced testing in the 2026.2 cycle\n\nChange-Id: I6f5bcf88684e50ab9941e93a63720b998f6a4e6b\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a1a05dcf4acd0749a3db64a4ebe50c4134c0742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a1a05dcf4acd0749a3db64a4ebe50c4134c0742"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
