)]}'
{"id":"openstack%2Fnova~1004781","triplet_id":"openstack%2Fnova~master~I44924d1f70ea3c05c8011720accb85e979a455ab","project":"openstack/nova","branch":"master","topic":"doc_missing_os_vif_ovs","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-09 12:07:32.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-09-09 12:56:28.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I44924d1f70ea3c05c8011720accb85e979a455ab","subject":"Add pyroute2 in doc dependencies","status":"NEW","created":"2026-09-09 08:23:07.000000000","updated":"2026-09-09 13:08:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94ec66c12ce16b069b6697b2b942911324cca6d1","_number":1004781,"virtual_id_number":1004781,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 12:50:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 08:58:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 09:18:04.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-09 11:37:18.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-09 12:56:28.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"886fc513703c84148dcaea89f069870cdc56ceb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-09 08:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8352c8776546b99269555a66705f441463dea6b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 08:58:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bf439dd1c68457cb27f71237eb874af\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/f0bc9b3e42cb4e74bed7749e28a03ac2 : SUCCESS in 5m 23s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec11d051714b4ded8f3a8288527ad95d : SUCCESS in 3m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cd365a2f80264c658a4a81c2d0105ddf : FAILURE in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98e0c6d0e0a4fba876fb293706c8677 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98cddc23a18f4e1aa403335c51a7514f : SUCCESS in 15m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0fb93b4b4524943b83fa60c0499868d : SUCCESS in 8m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6001330fae0042f3a74383410c002ccb : SUCCESS in 4m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/43f0b92b3b2749f9aebbe0f8a38e0fb0 : SUCCESS in 24m 56s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4c9b5e8b4dca470687397cae2acdfafb : SUCCESS in 29m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/83c0d6d30c6b46f2a54ca3818d89573e : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e5d04078960f0893a2a6feedfb3e4584f132d9f9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 09:18:04.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I44924d1f70ea3c05c8011720accb85e979a455ab .","accounts_in_message":[],"_revision_number":1},{"id":"95ac7f35cacefa0a77a59f2e218f4a8e41a0c4ec","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-09 09:40:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"181b2c3f4975835e2420302e69b0116ef2831256","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-09 09:42:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"351303d89b9a3d619ba587290c86916834f03802","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-09 11:37:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b35a71955747854129dee5fa67306d21bb6aa9a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-09 12:07:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7495ff95f16c1651a463af78cbbd32a23e89ea61","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-09 12:07:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50d9cf4257019247b4ecc4218c6c5fce7509c213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:50:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be1c4973336841c4a4ce5d63f98144e4\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/c2b875b1fcbf497fafc40f97d2ea8de8 : SUCCESS in 3m 41s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de3caf7a585440bdafee1d6fb54738e4 : SUCCESS in 3m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d57a26309f7d4b99b7ad9361e6859bd1 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0b48c3161c6410dadef0a622cc74f2d : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c2ab9eafab491d8074b5199a3745f4 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/545c0fb0dd9748c6ad2ad5b0c2e10e42 : SUCCESS in 8m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/91ccaf82b18b4d179b29cb57f36dc9ac : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6514c62af20640d8b9f8ac214a7bf58b : SUCCESS in 28m 54s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/379a0d60f57c49b48b46988ed3f8542e : SUCCESS in 26m 39s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f6dff421176f4bc18da3e7623d72e7a8 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1ee830b7f0e08ffd30f942f06818e60f8cba590c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 13:01:55.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I44924d1f70ea3c05c8011720accb85e979a455ab .","accounts_in_message":[],"_revision_number":2},{"id":"94ec66c12ce16b069b6697b2b942911324cca6d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-09 13:08:54.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"afff69a4749f53280b260709a060d011ebaba636","revisions":{"ed3c37f2d4ded188a1043cd6b88f6ecc41bba9ab":{"kind":"REWORK","_number":1,"created":"2026-09-09 08:23:07.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/1004781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1004781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1004781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1004781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1004781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1004781/1"}}},"commit":{"parents":[{"commit":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","subject":"Merge \"Parallelize per-node resource updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7582db6d65b9880e1ef7d906ddad1d7c95bb53a5"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2026-09-09 08:20:40.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2026-09-09 08:20:40.000000000","tz":120},"subject":"Add pyroute2 in doc dependencies","message":"Add pyroute2 in doc dependencies\n\nGeneration of doc rely on install os-vif to be able to load config\nparameters but this is currently silently failing in gates with:\nWARNING:stevedore.named:Could not load os_vif\n\nAs a result, the published documentation is lacking all parameters\nregarding os_vif_ovs\n\nThe issue appeared recently because 2026.1 doc is still ok.\n\nChange-Id: I44924d1f70ea3c05c8011720accb85e979a455ab\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed3c37f2d4ded188a1043cd6b88f6ecc41bba9ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed3c37f2d4ded188a1043cd6b88f6ecc41bba9ab"}]},"branch":"refs/heads/master"},"afff69a4749f53280b260709a060d011ebaba636":{"kind":"REWORK","_number":2,"created":"2026-09-09 12:07:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/1004781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1004781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1004781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1004781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1004781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1004781/2"}}},"commit":{"parents":[{"commit":"7582db6d65b9880e1ef7d906ddad1d7c95bb53a5","subject":"Merge \"Parallelize per-node resource updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7582db6d65b9880e1ef7d906ddad1d7c95bb53a5"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2026-09-09 08:20:40.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2026-09-09 12:07:20.000000000","tz":120},"subject":"Add pyroute2 in doc dependencies","message":"Add pyroute2 in doc dependencies\n\nGeneration of doc rely on install os-vif to be able to load config\nparameters but this is currently silently failing in gates with:\nWARNING:stevedore.named:Could not load os_vif\n\nAs a result, the published documentation is lacking all parameters\nregarding os_vif_ovs\n\nThe issue appeared recently because 2026.1 doc is still ok.\n\nChange-Id: I44924d1f70ea3c05c8011720accb85e979a455ab\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afff69a4749f53280b260709a060d011ebaba636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afff69a4749f53280b260709a060d011ebaba636"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
