)]}'
{"id":"openstack%2Fnova~1005336","triplet_id":"openstack%2Fnova~master~Ice8c1d6e724836253c98e3a18517069d4169368a","project":"openstack/nova","branch":"master","topic":"bug/2167088","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-12 13:13:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice8c1d6e724836253c98e3a18517069d4169368a","subject":"Make tox cover non voting to unblock RC1","status":"MERGED","created":"2026-09-11 18:50:42.000000000","updated":"2026-09-12 13:15:49.000000000","submitted":"2026-09-12 13:13:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005336-bug/2167088","meta_rev_id":"79c22bc84fc2bf15527bd4ec831000ce3d91df93","_number":1005336,"virtual_id_number":1005336,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-12 13:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-09-11 18:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2026-09-11 18:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2026-09-11 18:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 18:53:18.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2026-09-11 18:54:13.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2026-09-11 19:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-11 19:46:41.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-11 20:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 21:02:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"da7815e06bbefba69083cfb4ac1416adb2132c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-11 18:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3471bc00e68120f4cec2ffb21849a468f1a4636","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-09-11 18:53:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f9afb01ad4d41b352cfd488dba8ad296901de47","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-11 18:54:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1e0eb35622ab2ab2d6aeae0208466d1a668a788","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:11:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10961f75947945f7bc9c9e78d4a32a21\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f8333f3635c4230b26d05723eccdd4b : SUCCESS in 17m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9e01477d48b4206a7c8db3ad3e481ff : SUCCESS in 15m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5263420b13b04592902dcb7a3182cb62 : SUCCESS in 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66a567c8978663cd2fa131042ff89c2faf67c4b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-11 19:46:41.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ice8c1d6e724836253c98e3a18517069d4169368a .","accounts_in_message":[],"_revision_number":1},{"id":"22d9aaa6418ed77c242b74131b1079f0ee98fd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4e1a23f3a14cd582b4f1cfafd63a9e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61f4a4a693fd4d3eb1c895e29f209167 : SUCCESS in 1h 03m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/da3180691a2543ab9dd95395336aa11a : SUCCESS in 1h 32m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ca8251432b448e088258d870f84b400 : SUCCESS in 42m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc896ebb6c70491487dd7d7c67cfa900 : TIMED_OUT in 1h 41m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde475ed13ba4aa8ab776c3bd189e6f6 : SUCCESS in 9m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69eaf6c0ed5c45c682444ea197103e3e : SUCCESS in 13m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f4b8fb430004ca1a2f0dcd6143fdb28 : SUCCESS in 13m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50c71022297349c9bb08d33519ed5e6f : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11d36eec7314282a2067291ab16a669 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e72d5621f74c5fbb6a00af7ff96b8e : SUCCESS in 6m 08s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/13d235efc0bc429fa5011a440fcaf670 : SUCCESS in 1h 12m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d8e7d9a0ab54e82a6b10e35d880776a : SUCCESS in 1h 30m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6d6dc0e8f635406db5bd670108c8e676 : SUCCESS in 37m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/de201a43a70842c1a85352f34a0f5141 : SUCCESS in 36m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/12eb279393d34094ba47924d093335dd : SUCCESS in 45m 17s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/180e3706b5154b6a9e7c3f62d17afadb : SUCCESS in 52m 20s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b25c1952e41b4f10a992ff4a22080a1b : SUCCESS in 1h 03m 18s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/53b1107a1d0a4bec86855370b10de47c : SUCCESS in 1h 59m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/660236d1e674473ab82574d6b74309b8 : SUCCESS in 2h 00m 43s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/024efad4dfe74cc8aec95712d7abcc19 : SUCCESS in 38m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b068714c9d9a4272a3c710d12227f152 : SUCCESS in 5m 33s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c526a2294ec94f129ef32023910d0d7f : SUCCESS in 29m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1e65cc982b5b45adb2f2d38cadd9ae74 : SUCCESS in 28m 23s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e253672cbdaa4003ac3d415d12a6062e : SUCCESS in 31m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca13a6cee7a04d49958596d9f5d0e84a : SUCCESS in 1h 29m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51dc1c9147cf4036acb53ba6eff6955d : SUCCESS in 1h 06m 02s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/78d3398d028c4bab9a125719162a0f4e : SUCCESS in 56m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2b82d6a665f6410697ded885d2154138 : SUCCESS in 34m 44s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/af713ce00a9e4397a59259c4906f2237 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d1e85b6afeba5d7ef5e047bf3423499612a0e1c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:58:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a5448291a96257e2e58621839eae93967b3d784","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-11 21:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"bc270f6b9c19446ac82d98c23400c62a1bd48f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 00:24:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90446ac6b3244387a7ec8317f6722a8b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26fbea0ffb4144a28f473d34fff2eaf7 : SUCCESS in 1h 00m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e9c1a073a61b44a0866fe7de16cb79f4 : SUCCESS in 1h 31m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/232de1a465b04efcb0b0e55c16db7afa : SUCCESS in 57m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54cc8db9f4f4adb97e93b5f077c2b94 : SUCCESS in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4da72a2f9b1e46e0829731c8a9f9ce13 : SUCCESS in 13m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5a1a2f989a342279703fe513611055f : SUCCESS in 13m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa2afa695214321a5246613415323e7 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb9862e4c29643028def4659ae44798d : SUCCESS in 6m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e9a63f7f4a5b46f1b53d7b1d489ff741 : SUCCESS in 49m 56s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/eb3f7faa60894f1fac2adaf89f2dad8d : SUCCESS in 29m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/88610adc5ff6477b8a54ae79c29655e5 : SUCCESS in 24m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/cf7e2f3a93c3486283381f3d3399e72f : TIMED_OUT in 2h 01m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/913f9d80238b408cabbff5267faec056 : SUCCESS in 1h 00m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/7ac992b811304ec1a0fb00b48e046ef4 : SUCCESS in 2h 06m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/00023c4a9e564b36afe2421e85a4ec81 : SUCCESS in 5m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad927bfbc53f431fa6b76340b00dc731 : SUCCESS in 1h 32m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/41fbd63b7d6c40859d73d4883e3dd12d : SUCCESS in 53m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f42bbc647a6544fe97ceeb8f8ca2d258 : SUCCESS in 57m 03s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c42dd6a087f443b08f594d5c4b47bb5a : SUCCESS in 14m 36s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/15de56f3e1364ea8a425e590f3febd72 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"afc641830d7603a2a64eba2a888d3321a7a51223","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 06:58:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b125e748ceb2f04cd3c275c2fce01d56912c513","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 07:34:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2065d22ff0d4a87abc011c6c254201e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dae07fd6d480412e895918f6e28bed99 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03a44a34a0f14c58ae7bb59476653ff2 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1427709d324d4f8c9635286d62fa3bc0 : SUCCESS in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7093b586749d6576b5d14fb10e2b6cff57d4b220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:01:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6a06ac4935146a7bed7edb8bb44d818\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14f5f443395c45809eac1888f5a958fd : SUCCESS in 1h 03m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/feb66632e1264175889e46a170567418 : SUCCESS in 1h 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be900cac1d15420484120167439ae810 : SUCCESS in 1h 05m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4430900b46aa462dbfd37743c0b7999d : TIMED_OUT in 1h 41m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4dc5c2a700442c9325d3799581460c : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/435e7619e0434bc4afa8e3abfd2d5339 : SUCCESS in 14m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e01f230730e46d39e0daccfa1910eb3 : SUCCESS in 11m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2eaf1d531d44cb8956c40ebf0957b4d : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c5d14f492a42ff88aa318f63490efd : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9d2657e39e4423dbd9a3482a026df99 : SUCCESS in 8m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1f2d1d2d847c45479de586b5cccc5227 : SUCCESS in 44m 35s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/84767598ef5a477f9b11bfd91c5fb8b3 : SUCCESS in 1h 36m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/349a16e480bc44b399c1da7b618d636c : SUCCESS in 33m 48s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ce551e0e256245dea6c1ae32f0b28f75 : SUCCESS in 36m 48s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6c2837e8cc1c4253b446f072f98cd203 : FAILURE in 47m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f37b5cd7ee004c3d8e2bf2828d3145a9 : SUCCESS in 54m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fb4aa79eba68475e8f5167eac5a12c2b : SUCCESS in 1h 10m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/65a853332295433394ae2b980eaf1136 : SUCCESS in 1h 34m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/64951e0c45ff40ba88380f86f13f5fb9 : SUCCESS in 1h 48m 27s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e0a6f6f007584994b8c4a4e5790c69e0 : SUCCESS in 47m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/345041baa96c44a8b87848fd800b1e05 : SUCCESS in 4m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7b1800a2d2b448a18a5b4e4972970c1c : SUCCESS in 28m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/56bdbf79efd448d69eb5401e9af80cac : SUCCESS in 27m 10s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/04db54d507824cfda5a0d1a4f22cf3f8 : SUCCESS in 37m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d405676ed034f2b9c258a48b9f8646c : SUCCESS in 1h 28m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e60ed27efadb43f09e0668f29977a1d7 : SUCCESS in 57m 32s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/69622ae01d0a4483970d6db1552b3e44 : SUCCESS in 40m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8f724563f3da4662aa885a3aeec8298d : SUCCESS in 34m 27s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/af9fffcd6a5040eeb6db2de490d62c1c : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"752d7783b43ddf77bc3d3e7f4d9d8eeb941c6a54","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-12 09:30:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"876b3794d1946dc5c79b52b963ede3c29c6ea65e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:52:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af369d9e51d6444f9cc1504e0bea085c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0f959ace35d4466e8d7a29f655a8c694 : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25c19389222a4b2c84dce11a67febe76 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7f339ff46a4f4011a06b2854199eefdf : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e904240e3e91dbe3f46f1ec8a261f31ecd150a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 11:13:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a0c6bb963449efbebf7df3cbb3fcd7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3619e8d1552405eb39406fa52d9f1f4 : SUCCESS in 45m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7c8e976277e34eb5baaae53a4f318376 : SUCCESS in 1h 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e683a82ca82415f9bf5fcd8966cca82 : SUCCESS in 40m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd1d9ded19ca4e7594cc2a2d3b50b812 : TIMED_OUT in 1h 41m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba34518657d74b9db3e13012dd1f6e4c : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd0d650b71b8439f98c2e593f95a9334 : SUCCESS in 14m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/495dadadfdb345f7b766061bd8ec61a8 : SUCCESS in 12m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed0f65ffda724bec890b5a06236b6f61 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb737f2a01584e4c80c65341f3585479 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5103088a93684e8aad0aae39c7fa0160 : SUCCESS in 8m 36s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b4d7a7c6f1344bb9b3859765c82b7ca3 : SUCCESS in 1h 10m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36d8a207be0e495ab4973aa8f3151219 : SUCCESS in 1h 24m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fb74226fca8b4d2ebf260a98ca71cb96 : SUCCESS in 52m 00s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/16a6f26b7378474c86226616ea5fc928 : SUCCESS in 37m 24s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/91b0c92f6f474dfdbce0cedcf030f71b : SUCCESS in 48m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/04f95b893a1f464aac86ef1691012707 : SUCCESS in 52m 40s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b4e8052ab0e2414bae3d59d35c781029 : SUCCESS in 45m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/57de25f47a234365877718ea6e2d4ab0 : SUCCESS in 46m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/0f0207bff899475b867d65fec11efcfe : SUCCESS in 1h 35m 31s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/354a4f97e9524c019aeabbc81fa2e0fd : SUCCESS in 31m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b4ce6a1fb51f42b691724655a095a084 : SUCCESS in 5m 35s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c8e18e44553d4862b4f030e8304dbf32 : SUCCESS in 32m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e42d5947645c403abd3fbe27bc714c21 : SUCCESS in 27m 03s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/96622837caba42328c64ba20bcd1250f : SUCCESS in 36m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cf7a4056b7a41bdb1c0e6bcee544b58 : SUCCESS in 47m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c72e37a0202643a79295b44883a83978 : SUCCESS in 54m 38s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec90548975624a77ac5eaaf5cb88c719 : SUCCESS in 41m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/564d48e2d9914701bb46e2a62078d269 : SUCCESS in 32m 32s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3cad7546377c4826b05f1a77b10eef25 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b02ebf46dad814ebf15ae79dbe0386263f7a1ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 11:14:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c04d6161df46a5edda7e23056c65aee43ee253b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-12 11:22:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"9c86e680dbe9284665052a0d7e60d20b6a187720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:13:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b124be2c80e04a079dd90305aad9b937\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b869c74085614dd1bcc6b6093926fd28 : SUCCESS in 1h 05m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/262e67329f5141d29cb896c08aed6212 : SUCCESS in 1h 42m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e5c1a28c2c148969bbc1c6d13dd0ec8 : SUCCESS in 1h 03m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531a779c40ae4aabbd3210b4ded04b79 : SUCCESS in 8m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d2c226078c44c389b39fca9f70ab373 : SUCCESS in 15m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b62a181cbb7d4c9086f91a95b2461890 : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13bc5756fdab46c68599471541a44c78 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb1e9fc7c35146269791bc5567b22221 : SUCCESS in 8m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2b70123097fe48b39f526eec39de22bc : SUCCESS in 49m 41s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/46993da63c5040f08f234a65adc6b6a5 : SUCCESS in 21m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4dce2b8a8be7437081290967dae14e02 : SUCCESS in 26m 46s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3fd130c062c44444a170a81b7734f9a6 : SUCCESS in 35m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d2f0c8beefc7433291c9e8a4774ef485 : SUCCESS in 1h 22m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/1caac85c74b045d28d7278974b82cd66 : SUCCESS in 1h 52m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/61e48851b0054c539cab336a4a83d75d : SUCCESS in 4m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d38c427122540adbda3576c0b3ae75f : SUCCESS in 1h 14m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f00dca5a33354ce3b1719c0d9afdee58 : SUCCESS in 1h 26m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c28e538b78144f3b03a3ba3add430b0 : SUCCESS in 1h 01m 37s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/38a5d96c53134db5ab28c6519a62e5f0 : SUCCESS in 10m 37s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6ead34ab3dc543d28a0ebe488000cc07 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"319fd0334a744e5afd46e67e4e352a2eca186f76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:13:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79c22bc84fc2bf15527bd4ec831000ce3d91df93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:15:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3542c037743d4ee0a7f0c32b109dc520\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee203a257f2849689cb6964d2217e1e6 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b23eb491606eee2b3ff1e317860ecad65e6cccb3","revisions":{"b23eb491606eee2b3ff1e317860ecad65e6cccb3":{"kind":"REWORK","_number":1,"created":"2026-09-11 18:50:42.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/36/1005336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/36/1005336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/1005336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/1005336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/1005336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/36/1005336/1"}}},"commit":{"parents":[{"commit":"6710f10a0aa8a8fad2f03d865f0d1090044dab3c","subject":"Merge \"This adds an alias for Hibiscus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6710f10a0aa8a8fad2f03d865f0d1090044dab3c"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-11 18:50:00.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-11 18:50:00.000000000","tz":120},"subject":"Make tox cover non voting to unblock RC1","message":"Make tox cover non voting to unblock RC1\n\nRelated-Bug: #2167088\nChange-Id: Ice8c1d6e724836253c98e3a18517069d4169368a\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b23eb491606eee2b3ff1e317860ecad65e6cccb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b23eb491606eee2b3ff1e317860ecad65e6cccb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
