)]}'
{"id":"openstack%2Fnova~1005513","triplet_id":"openstack%2Fnova~stable%2F2026.2~If74aa8a00b851077ca159bb8e7c84ca34196d75f","project":"openstack/nova","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"If74aa8a00b851077ca159bb8e7c84ca34196d75f","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-14 09:10:59.000000000","updated":"2026-09-14 12:08:16.000000000","submitted":"2026-09-14 12:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005513-create-2026.2","meta_rev_id":"da6100bde723128f1ed0497756e119b5a06a42a5","_number":1005513,"virtual_id_number":1005513,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 12:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2026-09-14 09:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2026-09-14 09:48:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 09:48:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-14 09:56:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 10:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:58:37.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"df2a288c1a372afb25ac355a2d8a78b0d04b2f35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-14 09:10:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6841bd81c14a902fdbcbfc9513c8820ed17e8366","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-14 09:48:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4bece9185351ae7f05bcd4378fb1250a36026d7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 09:56:11.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/If74aa8a00b851077ca159bb8e7c84ca34196d75f .","accounts_in_message":[],"_revision_number":1},{"id":"6f54be0c21ff2092ef6772b09cfc4f5cc404ef30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 10:58:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3e981ed6d04ff3a6de5d043981ba63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55561c15eb494ab688eff183693886b8 : TIMED_OUT in 1h 41m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab582701603740b3bb774dc4ae263a52 : SUCCESS in 9m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8dce37dcffa6407d834548b93b0dc794 : SUCCESS in 14m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d06ce6b35e7947bda7962f138b32c174 : SUCCESS in 14m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/acf23650b96c40d7bd22c1527152345c : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781b36700dae471387d4880eebaafdc8 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ba9d986f9d844ac97bc7f13a9da7c97 : SUCCESS in 8m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a7cfa5a81dc946bfa60a3477fa0b7563 : SUCCESS in 6m 56s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/33b378fb845f4e2789b07e61883c4cf1 : SUCCESS in 26m 39s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e07196b75269469a92f2efabe3b0dbaf : SUCCESS in 27m 31s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8c410e24cdb148e7b1b5b019b5f65d76 : SUCCESS in 33m 22s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9228f12195e34326b82d24ab69dcb9d8 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8564cb2faf16ca89f1ca204f046f7ab5de776de6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:01:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2faa331bedc4bb4f52a6027adf27b401f6991fb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-14 11:07:53.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"02ca6a79b88a765348e98155c1c08d626658fd13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:05:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b7466fbae94bdfb099dc8ec0af1cbf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b9a5e78a844689862be8b39adb2260 : SUCCESS in 8m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8651cdcf89a740f291d314d207e2c9aa : SUCCESS in 15m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c32df92513b4a08a22fd0b8a7ff7d54 : SUCCESS in 14m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff648053e52a4c30aa6856ea9802e2be : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcb967d8bace4d15b5a8a0ffeec1074c : SUCCESS in 10m 07s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a4409a67ce0f448fa15df3f1ff41035a : SUCCESS in 28m 15s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/52ce3d7a3a63450eb497d080736a44ef : SUCCESS in 30m 19s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7ee75078ce714f40b3cb0f38e0e4f92e : SUCCESS in 37m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e7ec504cb87f47b8a46850ac496c1938 : SUCCESS in 6m 40s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/25ce80d62eb843f9a0f54f64c7a77e90 : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c0d40d566c723db982e01ab875cfb602b283f0ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:05:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da6100bde723128f1ed0497756e119b5a06a42a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:08:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3428895162af4e808edd85dd71b3623c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4708aca1f30643859791579d2e313f87 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbad2f3434424b5a961ea17e6e64b92c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcae7ceddcedfe2c0e623cc8ab3fcad8690ba1dc","revisions":{"dcae7ceddcedfe2c0e623cc8ab3fcad8690ba1dc":{"kind":"REWORK","_number":1,"created":"2026-09-14 09:10:59.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/13/1005513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/13/1005513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/13/1005513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/13/1005513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/13/1005513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/13/1005513/1"}}},"commit":{"parents":[{"commit":"8e8e9fe4d91fe15363985a1ef84ed1ea90cf0b3e","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e8e9fe4d91fe15363985a1ef84ed1ea90cf0b3e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-14 09:10:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-14 09:10:52.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: If74aa8a00b851077ca159bb8e7c84ca34196d75f\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dcae7ceddcedfe2c0e623cc8ab3fcad8690ba1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dcae7ceddcedfe2c0e623cc8ab3fcad8690ba1dc"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
