)]}'
{"id":"openstack%2Fnova~1005587","triplet_id":"openstack%2Fnova~master~Ifc8fe88d3cb6a1d6cdace41fe5a5ebc28080b589","project":"openstack/nova","branch":"master","topic":"prevent-orphaned-attachments-on-preempt-delete","attention_set":{"38902":{"account":{"_account_id":38902,"name":"Daniel Schulz","display_name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","username":"schulzda"},"last_update":"2026-09-14 15:45:30.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ifc8fe88d3cb6a1d6cdace41fe5a5ebc28080b589","subject":"Add regression test for https://bugs.launchpad.net/nova/+bug/2167250","status":"NEW","created":"2026-09-14 14:51:28.000000000","updated":"2026-09-14 17:49:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2d326183a5e3c3df6695eb1e455792f79aab380","_number":1005587,"virtual_id_number":1005587,"owner":{"_account_id":38902,"name":"Daniel Schulz","display_name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","username":"schulzda"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-14 17:49:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 15:24:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 15:36:17.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-14 15:45:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c5a286f08e204f3fc9ca3d15ee2cce1e76d5ff4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38902,"name":"Daniel Schulz","display_name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","username":"schulzda"},"date":"2026-09-14 14:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bcbb706b20983976bb7daa7b60c6315ec3646c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 15:24:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca09a4cbfbee44acb5afc53fd026fa4f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b2cba064ec544d88b4b65bb56347ed3 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5ef04ee553064852b36442d50e6ac573 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f616425ecc9f42c5aa018f0d567a8834 : SUCCESS in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a90714bd5c4fe935555fe31f35bc4ef3c2134cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 15:36:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ifc8fe88d3cb6a1d6cdace41fe5a5ebc28080b589 .","accounts_in_message":[],"_revision_number":1},{"id":"0abef46907d38980b41bcef50ab638f0b8653fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 15:45:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/380eab8bade8401f9461b581b15394ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6fabbc859b54ae784309f51dd39617e : FAILURE in 52m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6f9a7865484e07b403b256238721b6 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/822512b4494a48ea9e7cac3937179a7f : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7579eebd27ed4a16952874727d6a33ae : SUCCESS in 14m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10010469ad014980ac5136895eb51cdb : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ad2f8f701b43e58992cd4f92d6865f : SUCCESS in 13m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed036830c65406cb960376ad6edc4bc : SUCCESS in 8m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c4f08bd9b84b49da80a2bca5dfe448ca : SUCCESS in 3m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/426f1644481644238fbaffc631f87e26 : FAILURE in 28m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/411189cf6265466fad3e7561b1b428a7 : FAILURE in 24m 28s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d66f97de2ad0448c83bea1728592c3b9 : FAILURE in 37m 48s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/972f9b31addf4d4c9973517b4ad06ce0 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"005e6112b1c67576434d528093f59bbbd2925892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38902,"name":"Daniel Schulz","display_name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","username":"schulzda"},"date":"2026-09-14 16:01:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7bfa283d0259764f86315434eeefaec1fab46344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:27:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d15effac45b4b83b782578dd80dcf65\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d15574fc5742408483b45a37d3db140f : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2cfd81a01ad4e178d3250512386de21 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a7f3caf26f5a4783bc1f346ad3662bf3 : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f11bcf7b7ba0a85a0931669ca4c0f3a614f5839e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-14 16:45:01.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ifc8fe88d3cb6a1d6cdace41fe5a5ebc28080b589 .","accounts_in_message":[],"_revision_number":2},{"id":"a2d326183a5e3c3df6695eb1e455792f79aab380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:49:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2102bff83a646cb8a4dfe6fa47b401a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd0fb0c928445e49645ce780a3b749b : TIMED_OUT in 1h 41m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6654d9f243c143509368ba9a7ceeced7 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de9bfa49dbf14d108b3af9bc395bf7c1 : SUCCESS in 12m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4c694be5a1c402f9b88035bfc9bb513 : SUCCESS in 15m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3fb85bad98764830a3ec511a75196499 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e1e542af484b299ad7a9c1a8649d91 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b763d3b2f964f9aaf2319433c7b84d0 : SUCCESS in 4m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a501a3bbfb9c4948a73f71b39d23b6e2 : SUCCESS in 5m 22s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ef137c2093e1469abb98b2d8f878d3bf : FAILURE in 27m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d4d26ba5fd4640db91dcb3aecf3b4e64 : FAILURE in 23m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a32498cd03014e738e99e0dd92e03c9a : FAILURE in 30m 02s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/52b53d579c4148fba7b5d17afc45fba6 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fbe04daaf693cfbd4c93dadfd78a966db2ba9217","revisions":{"0ea87632768f684efb033f9ba03090acb2b8e701":{"kind":"REWORK","_number":1,"created":"2026-09-14 14:51:28.000000000","uploader":{"_account_id":38902,"name":"Daniel Schulz","display_name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","username":"schulzda"},"ref":"refs/changes/87/1005587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/1005587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1005587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1005587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1005587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/1005587/1"}}},"commit":{"parents":[{"commit":"9aa9a54e04fdfec9a8775bf90488558783179d4b","subject":"Merge \"Filter out deleted instances when looking for build timouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9aa9a54e04fdfec9a8775bf90488558783179d4b"}]}],"author":{"name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","date":"2026-09-14 14:47:14.000000000","tz":120},"committer":{"name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","date":"2026-09-14 14:49:10.000000000","tz":120},"subject":"Add regression test for https://bugs.launchpad.net/nova/+bug/2167250","message":"Add regression test for https://bugs.launchpad.net/nova/+bug/2167250\n\nRelated-Bug: #2167250\nChange-Id: Ifc8fe88d3cb6a1d6cdace41fe5a5ebc28080b589\nSigned-off-by: Daniel Schulz \u003cdaniel.schulz@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ea87632768f684efb033f9ba03090acb2b8e701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ea87632768f684efb033f9ba03090acb2b8e701"}]},"branch":"refs/heads/master"},"fbe04daaf693cfbd4c93dadfd78a966db2ba9217":{"kind":"REWORK","_number":2,"created":"2026-09-14 16:01:40.000000000","uploader":{"_account_id":38902,"name":"Daniel Schulz","display_name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","username":"schulzda"},"ref":"refs/changes/87/1005587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/1005587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1005587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1005587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/1005587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/1005587/2"}}},"commit":{"parents":[{"commit":"9aa9a54e04fdfec9a8775bf90488558783179d4b","subject":"Merge \"Filter out deleted instances when looking for build timouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9aa9a54e04fdfec9a8775bf90488558783179d4b"}]}],"author":{"name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","date":"2026-09-14 14:47:14.000000000","tz":120},"committer":{"name":"Daniel Schulz","email":"daniel.schulz@digits.schwarz","date":"2026-09-14 16:01:08.000000000","tz":120},"subject":"Add regression test for https://bugs.launchpad.net/nova/+bug/2167250","message":"Add regression test for https://bugs.launchpad.net/nova/+bug/2167250\n\nRelated-Bug: #2167250\nChange-Id: Ifc8fe88d3cb6a1d6cdace41fe5a5ebc28080b589\nSigned-off-by: Daniel Schulz \u003cdaniel.schulz@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbe04daaf693cfbd4c93dadfd78a966db2ba9217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbe04daaf693cfbd4c93dadfd78a966db2ba9217"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
