)]}'
{"id":"openstack%2Fnova~1005879","triplet_id":"openstack%2Fnova~master~Ibbe9a4597ee7ca19ccccb5b283525b2782dfe88f","project":"openstack/nova","branch":"master","topic":"bug/2167481","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-16 19:07:42.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-16 17:19:03.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ibbe9a4597ee7ca19ccccb5b283525b2782dfe88f","subject":"Reproduce bug/2167481","status":"NEW","created":"2026-09-16 13:36:45.000000000","updated":"2026-09-16 21:14:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d07f618d096e8bc6246b94ad337183101d18a8e7","_number":1005879,"virtual_id_number":1005879,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 19:06:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-09-16 17:19:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 13:50:58.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-16 14:34:59.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 15:29:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 19:07:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"837007598e71e19675a7a83c3603284f88300661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-16 13:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4edda00cd06e9cd0c8f39cda5717bf75ab3c0d5a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 14:34:59.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibbe9a4597ee7ca19ccccb5b283525b2782dfe88f .","accounts_in_message":[],"_revision_number":1},{"id":"fffa5cc8d3854deb09ad5faa0e85f8a5efd7d489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:29:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77a8a079ae5a42de98f33b4855d1105d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08ebb91629b143e38a111e1e04d67f19 : TIMED_OUT in 1h 41m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdbe40966731478587f8bc09b0415c1c : POST_FAILURE in 9m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ebed45451b04b7bb118e6668f9e70d0 : SUCCESS in 13m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31029e6c109240a1b27032c4e7b71d0a : SUCCESS in 14m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19cbae9c14504f4e9be512a5cbec7d7c : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efba773f2d0844c9863a88368a55cea5 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/130398279ecf42b5a552f1d5d5196e5f : SUCCESS in 9m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/93c0f87941f54285b95c612d83f80b74 : SUCCESS in 7m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6b10716569bb438fb069b359a27bfb22 : SUCCESS in 28m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1d33ec277b7d417f9f44b9b32e628395 : SUCCESS in 25m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/bd88cf3f3fbf4fdcb71783f4a11bbec0 : SUCCESS in 32m 19s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/077b35cd22054612baa8bb43649181a5 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4aa8dc6c2efb46489d8f66011240384be28a57d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:31:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9edc5520b19846cb96f55d8ec33476e5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15f234f9ed5141e7928ba1fad62764f7 : SUCCESS in 23m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bc1bb57822540c984a645d5c5ab6930 : SUCCESS in 27m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c7c0085c1b4048b88c357ccd0850806f : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30ad126e162c7cf3ca85abd9c1eab20b396017bd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-16 17:19:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f48b4e99b0ec5788adb07c54c83851500b544fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 19:06:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7d1b4c4e83f4a9fbcf14fa500548ca9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c134c2d421d84d3b960e2a88d5bc6cf5 : TIMED_OUT in 1h 41m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ddc7a4cbc2f44fc946c5e957f6e126b : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3db7690b38f5472bb32ad69480a5e824 : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c62341c7aade4988bce71a4a259e8a74 : SUCCESS in 8m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89e88e065bac4559be31293d45522587 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14579dffb46441659a3e9366346e5e8a : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c55876bd43bd4fc499ae5156937e8d5f : SUCCESS in 10m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/91b00f1e474c4e38b39c1bae14d19747 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/559d312939c844e29517637f2a4c6dc4 : SUCCESS in 27m 34s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/205d96409b6b43ea95474c1cee0a4c0b : SUCCESS in 29m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d7303160eea446d18046e08243a5517d : SUCCESS in 38m 14s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9c0d3f2d7266439f91a4ce7288fba7c3 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"20c31a237d1dbca8a110f7b8a434041f90a85900","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 19:17:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"d07f618d096e8bc6246b94ad337183101d18a8e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:14:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8e0ac1dce594098aacdc910268bda3f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01ef6799f9b349aea90c65564d4c2831 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/efab195917c84b299fbb88388099c439 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a4219925caec442ebfc6b69bf4415c7a : SUCCESS in 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d038741b80ca5f81e613894bdd9194281a414d7","revisions":{"4d038741b80ca5f81e613894bdd9194281a414d7":{"kind":"REWORK","_number":1,"created":"2026-09-16 13:36:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/79/1005879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/1005879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/1005879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/1005879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/1005879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/1005879/1"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-16 12:54:11.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-16 12:54:11.000000000","tz":120},"subject":"Reproduce bug/2167481","message":"Reproduce bug/2167481\n\nNova\u0027s logic to skip periodic tasks during graceful shutdown breaks the\nprotocol of oslo\u0027s periodic task interface. A task should return the\nnumber of seconds it should be triggered again, but our early out code\nreturns None.\n\nRelated-Bug: #2167481\nChange-Id: Ibbe9a4597ee7ca19ccccb5b283525b2782dfe88f\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4d038741b80ca5f81e613894bdd9194281a414d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4d038741b80ca5f81e613894bdd9194281a414d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
