)]}'
{"id":"openstack%2Fnova~1005880","triplet_id":"openstack%2Fnova~master~I44999b3354ae32f4cdf356b46d60d42e004e2ccc","project":"openstack/nova","branch":"master","topic":"bug/2167481","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-16 15:52:36.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-16 17:33:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I44999b3354ae32f4cdf356b46d60d42e004e2ccc","subject":"Stop periodic tasks during shutdown","status":"NEW","created":"2026-09-16 13:36:45.000000000","updated":"2026-09-16 17:33:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a2b8a240609177dc6e08498e1464a0614346d0c4","_number":1005880,"virtual_id_number":1005880,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-16 15:52:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2026-09-16 17:33:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 13:50:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-16 14:33:26.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 15:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70ff43bb5db865bd8b38c94b8cafeecb45234762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-16 13:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d4f0a410de3478274136b1dc73788ffe2cbb00f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 14:33:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I44999b3354ae32f4cdf356b46d60d42e004e2ccc .","accounts_in_message":[],"_revision_number":1},{"id":"29595f9d6cd0bf8eb09f59d53f0f4082138b8ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:52:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60449f87f5df4fa882668ae3ceb3fa0e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f62943cd401e4152a6226e6de215feb2 : SUCCESS in 54m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d34b260c93474b139db4d47063d8d153 : SUCCESS in 1h 53m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81b57f3bfc5b479abc4e5808dc450e3f : SUCCESS in 1h 03m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/426bd8380b5e410381fc1ef2bbd3b07c : FAILURE in 43m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb5872a1c274a26bfc1384bc4134174 : SUCCESS in 8m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67ea7d323edd4a0f903ce9fff524c02e : FAILURE in 14m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8677a83f967e47f6a74458c5a60145f9 : FAILURE in 15m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6ee80937a1144c1ab571bae26e16944 : FAILURE in 14m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3e59dda2b541e89f4760bc0022f006 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0698e56cab554f2fb472a33b9fccec26 : SUCCESS in 10m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/00a3110ef5304f778b471a004eee8cbd : SUCCESS in 1h 11m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6cd37189906643e48014d2d2c4667825 : SUCCESS in 1h 43m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/647916fc892d4ddeb308f62ad095ee1b : SUCCESS in 1h 06m 27s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/16f9cc0f0d0341d6bdb7fecccef3dd9c : SUCCESS in 41m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b4231253eea04485ae8583eae0c4d8a3 : SUCCESS in 1h 05m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/89fa429266ce461da85b506de914d58f : SUCCESS in 58m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2dd5d40788bc48c0b95e683e469f9435 : SUCCESS in 1h 47m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/e48edf7230b3404c99ec5e4f10bf122a : SUCCESS in 2h 05m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2f70bea788fc44ac8a0f1750ac02b36d : SUCCESS in 53m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fa5ea4e4b7914782af1c0a04cf41fadf : SUCCESS in 6m 04s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf4f03bc43244ad586a8da28f05268ef : FAILURE in 24m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/983f345535be4e69bfb4fdd95dcb02ab : FAILURE in 27m 51s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c2f80410cb484a408e48dc26eb0733c5 : TIMED_OUT in 2h 01m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/855c6f4dd26a40b2b7b6d5b1a3f9ff9c : SUCCESS in 1h 33m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b61a1337a77b403e9207fe759342449b : SUCCESS in 53m 58s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e56dbb0534ac4717b2f184cdc35fca85 : SUCCESS in 1h 01m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f795f81a042a42ee98f6901aee7ab306 : SUCCESS in 34m 04s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dbe97dae06844019aab36f1a2b5ceb6d : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8aee1fe7426e729a075fc18fe156720c2b333801","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:59:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16f404accf644841b392f6f8cab4c78f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8dc098c5c8c24bb8b1eca7b3d3719a43 : FAILURE in 24m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b389682451a4b6fbcdc0cfa70530a02 : FAILURE in 11m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/54bcba04f75e4694a46035deefb2148d : FAILURE in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2b8a240609177dc6e08498e1464a0614346d0c4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-16 17:33:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfbc62ebdb9a875bff41dfe1fbd8a319fdf9da65","revisions":{"dfbc62ebdb9a875bff41dfe1fbd8a319fdf9da65":{"kind":"REWORK","_number":1,"created":"2026-09-16 13:36:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/80/1005880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/1005880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/1005880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/1005880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/1005880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/1005880/1"}}},"commit":{"parents":[{"commit":"4d038741b80ca5f81e613894bdd9194281a414d7","subject":"Reproduce bug/2167481","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4d038741b80ca5f81e613894bdd9194281a414d7"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-16 13:33:31.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-16 13:33:31.000000000","tz":120},"subject":"Stop periodic tasks during shutdown","message":"Stop periodic tasks during shutdown\n\nThe oslo.service periodic tasks runner expects that tasks are returning\na number of seconds indicating when to run next time. During graceful\nshutdown we want to stop the periodic tasks but we returned None which\nbreaks the protocol and triggers a stack trace. Instead this patch makes\nsure we raise LoopingCallDone to signal that no more tasks needs to be\nrun during shutdown.\n\nCloses-Bug: #2167481\nChange-Id: I44999b3354ae32f4cdf356b46d60d42e004e2ccc\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dfbc62ebdb9a875bff41dfe1fbd8a319fdf9da65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dfbc62ebdb9a875bff41dfe1fbd8a319fdf9da65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
