)]}'
{"id":"openstack%2Fnova~1005881","triplet_id":"openstack%2Fnova~master~I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254","project":"openstack/nova","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-16 17:19:07.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-09-17 22:07:26.000000000","reason":"\u003cGERRIT_ACCOUNT_38797\u003e replied on the change","reason_account":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"}}},"removed_from_attention_set":{"38797":{"account":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"last_update":"2026-09-18 13:44:45.000000000","reason":"\u003cGERRIT_ACCOUNT_38797\u003e replied on the change","reason_account":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"}}},"hashtags":[],"change_id":"I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254","subject":"Enforce nullable\u003dFalse on instance_uuid","status":"NEW","created":"2026-09-16 13:39:25.000000000","updated":"2026-09-18 13:44:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf22794e4eef1df2d7f5ff58360a108e4e8e2746","_number":1005881,"virtual_id_number":1005881,"owner":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 23:53:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 13:49:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-09-16 14:41:06.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-16 16:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 16:25:16.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-17 16:09:11.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"7d6d55f7e35348683865c6d948766e182d570a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 13:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f751a500cdbdf312c8c71322f5a20f03a97111ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 13:49:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9d0f340e99193d69f8261b9f3604b975446ebf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 14:16:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ad292b1cfd9850a9b49769112fea6a28af098ac9","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 14:17:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aeedfa3ac207beb7d39c29e573b58d12871429fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 14:41:06.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254 .","accounts_in_message":[],"_revision_number":1},{"id":"c6840c7e89f9a8f27bbe4285ad78fd59840862a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 14:43:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fc58800daa2b44be65ce6bccbe329c0b1785605","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 15:14:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254 .","accounts_in_message":[],"_revision_number":2},{"id":"cc44d0b04a4f61bc3a86ff1e7b7306f254de02ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:25:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04efba4d0fc40a099b858ae722205bf\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f58b36785c2f4c00b3c55ba8a7e25ac2 : SUCCESS in 1h 20m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aa1c589b0ab64999a9d0acb4b5082369 : SUCCESS in 1h 45m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4b0369c93a54a0bb9d7faafdf870c6b : SUCCESS in 1h 02m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29173777f026432ebfae0c2ae0391c42 : TIMED_OUT in 1h 41m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c087fd07c004316958c6923b722f361 : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7375377df0144eeb968299b817a85d4 : SUCCESS in 16m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a34013392a5647898c9a335daf2b886f : SUCCESS in 10m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26dc0eeb534447c0a7758482c1a399af : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f814394898c54380958b5f66b0e24f35 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e108b59d1c4d98b233b29e50035c32 : SUCCESS in 9m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9f05a2bde55545f88b0480f89a353068 : SUCCESS in 36m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cdcb3c034efe4adabad6fedf0dcc944f : SUCCESS in 1h 34m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/93e8e1cc59644994b2146ba60cbe27f2 : SUCCESS in 59m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fac61b3b31cd43f397b261ed909c8e97 : SUCCESS in 39m 12s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9472f2f3c2f44b119379255bc2c55ffd : SUCCESS in 52m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7f5818eedf0741c6ba9756dbe3125a9b : SUCCESS in 1h 02m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e26b372df2d4e6d9576c87d275ec153 : SUCCESS in 1h 03m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/5086415a0bea405f8183bc3f92380df7 : SUCCESS in 1h 58m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/3a519cfb57c74a25a7caf1cafdf24f3f : SUCCESS in 49m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c3d3d25e0a74471b8a250dbc1bc24624 : SUCCESS in 3m 52s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/25c8079399d74068a292908b142df8be : SUCCESS in 28m 29s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bc6ca14d147f4281a6b24e5c128f22da : SUCCESS in 30m 57s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2271f388ccb14c89847738cd6f7b37a4 : SUCCESS in 36m 38s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f87482ec50eb4bae8c5446a9405e64ea : SUCCESS in 1h 16m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9624fcc05724bb6bc08e98194688e7e : SUCCESS in 32m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e1d7d05f7d1c4cc699901d83341b90f9 : SUCCESS in 1h 02m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/58869cbeef5b4d4a9f703d0d1751680b : SUCCESS in 41m 04s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6a75435cbdfa41bf9ee945221fb3ca47 : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"aebaa0869efe0a225aa0f9bf24addcb785d10cb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-16 16:34:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"5115f90440ae2eb363a44be6469e2549ef41c225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 17:19:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"371428a0c9ac0e5058561b0aee5fcc4a6afb5219","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 17:21:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3a1809fec1e7a8ea7823225207e10214e5106f8d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 18:13:46.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254 .","accounts_in_message":[],"_revision_number":3},{"id":"b0c3472de6987d14929bd11b9b2f0e5ec6927536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 19:45:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f467c5dc0a4f45818530893d0f5f54f4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/056ffeed260d4ee5b52cb11a6119e212 : SUCCESS in 45m 00s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/10bf1acb772344179328c0f79e01cadb : SUCCESS in 1h 48m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e83c234a4df4fb683c1683839667025 : SUCCESS in 55m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3967289e7fd24a97a6b6ddd3370686df : FAILURE in 30m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd148e51e024bfc90a30911f376e418 : FAILURE in 7m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ce749352ce94473b96c5261697ec4c1 : FAILURE in 14m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f301109c524c4a4ea2bca4cc46aa985b : FAILURE in 16m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/627910ca6ba4489394c93f752e4eaa74 : FAILURE in 14m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab49f1a063da4ceb928fab85c91db14b : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce656ae0b1ff4cba9957fce5b99875a9 : SUCCESS in 10m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d3c4a081b2d1480687ff133a0b0ddda2 : SUCCESS in 1h 11m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93427bc2fb26436397d85e7d722bce41 : SUCCESS in 44m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/972cd846a2ed4e87b1e6a558ea1851d5 : SUCCESS in 51m 29s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a9512901460543aca57dec620253677a : SUCCESS in 38m 13s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9cdab8e08f6c44a1a0029c095156be10 : SUCCESS in 59m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/802daada4a0a41a09e27e9a2dc082110 : SUCCESS in 25m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fa8245db75434b05b7325d26e7e1732f : SUCCESS in 1h 54m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/24af22fbcd1c407f8cd04561ef469f93 : SUCCESS in 2h 19m 46s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5055cadcda7f4c8daf46923b48fe807d : SUCCESS in 48m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4a5ea31749b84d74989ca49b823adc6f : SUCCESS in 6m 24s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5e5090c3941d4981986796ed71b9995d : FAILURE in 8m 49s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f8a5f7136ddc428698461ac35901ccee : FAILURE in 9m 27s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/46fb1989fe7a4ec4b03b40296ffca00b : FAILURE in 8m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/68168714b7bb4553a37973c1e2940843 : SUCCESS in 54m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d4f5feb75864c0dac4398ce8f1b300f : SUCCESS in 1h 08m 51s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6bca212de8f6441ca1d4da25eab7f27b : SUCCESS in 51m 29s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c5c53e0eeb2d4a26b1359334c4395c09 : SUCCESS in 15m 48s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/04786baa95b7485aaad1d2a4a080f70c : FAILURE in 13m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b09317a0b8855cb04fc4545b8f3349c244bb6a8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:53:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235ae3472f914204ba17e9bf7983a0e2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb23f542bcf942cd8e56362006ca2f35 : FAILURE in 34m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/97de6aa149794351950466545ce2652e : FAILURE in 30m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1e6bcd8fa1d648adb6b61f517c40d0bc : FAILURE in 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f0bb8932aa5e87af556fa12e7a987f3c3c65a6c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 21:06:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"2b118442f419e3378dd79b3f2b41eb125037c2ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 21:13:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b27a0e501e5dee94a37659e358b783c72b023cb1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 22:01:25.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254 .","accounts_in_message":[],"_revision_number":4},{"id":"fd4085c57990eb75a99b3e1805b38ff26ae9416a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 22:06:59.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254 .","accounts_in_message":[],"_revision_number":5},{"id":"3e584458a559415d3a04bd5dcb1ee3270c20a13c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 22:53:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1795a138ec4458c868caac74b1ff657\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0304ccd11414489b8dd615810fa2e9d9 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3d1dbb036eb45a08c4e6e0cfed7fbb2 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13996d1232944811a76c8e6c8d2fb048 : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"131fed6a9ca174152f9b00fa84306c3949480bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 23:33:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39084247f10e43279ce30d0a932afb25\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d18a7652ee84136998174348a896e13 : SUCCESS in 1h 10m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2f74a1d760a642e1815d86dd6668fe44 : SUCCESS in 1h 40m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7e1373884ad49db90d15fd2e6fd7d36 : SUCCESS in 53m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2028a7cb054b463d95b557f0f1c25964 : SUCCESS in 50m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7351318798f4fe2aaf9460b6a0c26b4 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c07a0770a84b4f999442173b8f69e5a3 : SUCCESS in 14m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c6ac69dde1145b786807b33f1740db8 : SUCCESS in 13m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b28ee5e7cad45c48cbf92d32caff9f0 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92485c5f55b1455bad1633aba44dfd20 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42ba437a0a2c459998955bd736a87aab : SUCCESS in 4m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/87b72401c3e34dfe9934e07b2a8645af : SUCCESS in 1h 18m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc2201deca1f471eb32826105257ba31 : SUCCESS in 55m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b8de1b554dca496d916ed096cb3e8732 : SUCCESS in 50m 13s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/77d6b41d6be34d5caa5cba5032e54e3a : SUCCESS in 36m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a75b97b242bd49848b58136b6aad7d11 : FAILURE in 47m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/17cb7cf6246e4b2dbe8e5914baf128c5 : SUCCESS in 32m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3f6ad16f5562472ea82a94c921ca8466 : SUCCESS in 59m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/c9653789c46c4d3f897af1646f66e0e5 : SUCCESS in 2h 04m 57s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/67d4a74194704f17acf2c2601d7492ee : SUCCESS in 49m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e0f62c3fbfbe491ca832f7931fc7e40a : SUCCESS in 5m 07s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a7109e63726d413d80502547afbdc284 : SUCCESS in 34m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/715ea64c668642c0b7e92068b1626939 : SUCCESS in 28m 03s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/79f185c089a94d2fbdbf0e5aaf7e6f2a : TIMED_OUT in 2h 01m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9729069787ba42709eaf31c3119170f9 : SUCCESS in 1h 02m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a468b441be814caab6c39e428c4389e7 : SUCCESS in 56m 05s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bb8121e6122849ca9bd91699a3e92463 : SUCCESS in 55m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ff0336921975437fab5e4f4ebcd8eac5 : SUCCESS in 32m 39s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f033666e266b4ebf8cf9f5c91413051e : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":5},{"id":"48fe7af9d1b266eee6c51d2d48966edaf4e3f5f2","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-17 12:20:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"806f4a3dfc9cd0f82a7da33b97f8041d23390713","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:12:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bd6504b5b204aac942d6a99d25d8a1b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2062489fb14146d29c54cdac8487e80d : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ced0f35db134426801ff9d029b0415a : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b9bfe9cbf1d84cbaa67c55c5a7dc8c3f : SUCCESS in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c9ef90c6562c1c7371d3356713bf176c4db322ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:33:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/723edf2dfdc948e689f7005f8bd212a4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ddbe00122edc46e19cdcf29af573bda3 : SUCCESS in 1h 03m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c03ce9ce54bc4cc1a78046a70fedaff8 : SUCCESS in 1h 32m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52dd9026bab1407b90000f4b8d02e793 : SUCCESS in 1h 00m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c799cd97cfa49b6b47666809a42ea9d : SUCCESS in 1h 04m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fcd680666c4f429be64b12f1b08705 : SUCCESS in 9m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8f031a2623f4ba6bb06447094f71397 : SUCCESS in 15m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecc4d10b6259401bbd06c024ba400aca : SUCCESS in 13m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6aac3d9ea47e462bbe221a0ea68d08ed : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b3968910ee4b8fbc807f8de35eff3f : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d42d46bd04ec4c90982c8c00cc9b0ee0 : SUCCESS in 9m 28s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e3f09b3c7be442d2bf430015bb6c94d6 : SUCCESS in 1h 09m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5420dd4b2c804d73ae65c67b45d4a607 : SUCCESS in 1h 35m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f8d08725e10f4874b375ffd717f9037a : SUCCESS in 43m 21s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f6a2fd8b40c74936a7b40789b13076af : SUCCESS in 41m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9a0cdc30a52f47f5954fa43bf9f9f4bf : FAILURE in 58m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bee0dec81cc3487b91f0d94c8f63e6a8 : SUCCESS in 47m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5dd0e5316c2a46a393b4a61cb111fa2b : SUCCESS in 1h 42m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/91a8d5dded4c412cbcfc8913d2260ff8 : SUCCESS in 2h 05m 26s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e84d810659e9460c99ccdf3b368aa47e : SUCCESS in 52m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/645268d513554c37a3613aa0c77dc3a6 : SUCCESS in 6m 42s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3d23b1d2aee94ab2b908df0dc5522213 : SUCCESS in 28m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e750a808a0984176a62a29b69bc7b324 : SUCCESS in 29m 05s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6fc6bf9f2c5f4148be34df6257c373fc : TIMED_OUT in 2h 01m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/60dc2b36decf49089acd66299185fbf7 : SUCCESS in 1h 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/599e1cb4f800498f89fdb4231cd03867 : SUCCESS in 57m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6ce1a4e2edf44c17b7f99d7e5185c453 : SUCCESS in 54m 15s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fae477ceb9f54e7f85df49a706c868ff : SUCCESS in 35m 20s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4189686276d240b98daab4c3bf952074 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":5},{"id":"4e5ae98ebc49f7d87dfb45ecfed7f1fedcd7c7f1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-09-17 16:09:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a6458e6dd79dd0ce3714e8135a9a647028cac0b0","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-17 16:28:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6fb2beaca19a40fbc1c143d386ef705ae3243437","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-09-17 17:16:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"350f6ab10ce85493946551c9e1e7a3fb05349331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-17 22:01:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"a5ea7af35019f0327c0ee7fdc3252512951dfe5c","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-17 22:07:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"314824769be75c1e7c3d003653344308314e5e35","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-17 22:07:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4bf449a9057b7a6bcb5b78ceb41ee356cda66895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 22:33:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46abd161b62b46e280d5f96d93b76bea\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/326d21ce42274f2080ad289628faacbd : SUCCESS in 20m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/434bd077324b4f078690bd7c818afbde : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51a3130c73f84feab8a804df17d008cb : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b8a5c349e70c9e7de07b1567f47e0830b91d06b4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 22:55:19.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254 .","accounts_in_message":[],"_revision_number":6},{"id":"01b5a85263bc1920cd5cc3faed61648ca9d68e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 23:53:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02d2b127b9d5438fbf713748e7082e5e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48140f769cab4a75aac16994f76a35dc : SUCCESS in 1h 07m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/94aed6a4933e4545a28ca103cbf42961 : SUCCESS in 1h 29m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db0b49cc9ab34e57a4f90dec369ca042 : SUCCESS in 55m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df9a666043044349966800a67a8d6d89 : SUCCESS in 56m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6f1902bc8c4cb496ff37fcd8a749eb : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb3857b8d82149b59b7d25f199be9f94 : SUCCESS in 15m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18d508b3fc254ddcb19bfa7c02b43618 : SUCCESS in 13m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bfd0235fbc344ee9a4f085ba0a4e6775 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db7b6a4f330426aa13917eab7dd4441 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0adbef0fa64f4dbfacde5ba03c438862 : SUCCESS in 10m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/61f2c0785dbe41718235cd6c6f360316 : SUCCESS in 1h 06m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a15fe3cfa284be18bceb658348dea62 : SUCCESS in 47m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/89be5a64cc1049b4b056f48caf65a62a : SUCCESS in 29m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f6a8375a5cec4b02aacbe55b9e50ebce : SUCCESS in 38m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c68d1c0414a5402681d6b1caf771e5d7 : SUCCESS in 1h 03m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f4e4b1dba62d49709c9c0a571fbea145 : SUCCESS in 27m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0b3ae618020a4bea89615eaf47e04378 : SUCCESS in 1h 43m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/2affee9b7a554cbdb8b05cd681acb881 : SUCCESS in 1h 03m 39s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/1e7f4ef32da1407ca3cadd7b37d310dc : SUCCESS in 41m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/92e5de19d90f4159a9467849bd9f82b7 : SUCCESS in 5m 34s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/954c5a6d4fa74884a36754c06fb8b0a7 : SUCCESS in 28m 47s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/10e8a7a3493c49bd8a19344359c865a3 : SUCCESS in 30m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1bff5215b82c42118152795332ae0ee8 : SUCCESS in 31m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/da3d387f6c08487aa5d460db73ab70ea : SUCCESS in 1h 25m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2b1d834132e4250b5e66fa6e6eda1b0 : SUCCESS in 58m 29s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a7bdbf917f3445ccb7e253c386ad63d2 : SUCCESS in 53m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d33f66d1c2fe4db28a80c6333512d17e : SUCCESS in 38m 34s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c47acc03181a4cc79ab21714d424d1bf : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":6},{"id":"3882e5e5dd6e07a0723ef24a45e82ec7705d74d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-18 00:02:41.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"bf22794e4eef1df2d7f5ff58360a108e4e8e2746","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-18 13:44:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f29e0c2c318105367cd4ad596a48f73293ea733e","revisions":{"6fb1dc30166f57bae97f922c287671702895ef9c":{"kind":"REWORK","_number":1,"created":"2026-09-16 13:39:25.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/81/1005881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1005881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1005881/1"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:28:05.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:39:21.000000000","tz":-180},"subject":"Enforce nullable\u003dFalse on instance_uuid","message":"Enforce nullable\u003dFalse on instance_uuid\n\nThis resolves what was requested in the TODO left by mriedem.\nThe instance_uuid must be set to nullable\u003dFalse because it\nis a core identifying variable that must always be present.\n\nChange-Id: I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fb1dc30166f57bae97f922c287671702895ef9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fb1dc30166f57bae97f922c287671702895ef9c"}]},"branch":"refs/heads/master"},"7a7a7711d0d15f4374c2998221ffaababa049e57":{"kind":"REWORK","_number":2,"created":"2026-09-16 14:16:44.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/81/1005881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1005881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1005881/2"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:28:05.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 14:16:27.000000000","tz":-180},"subject":"Enforce nullable\u003dFalse on instance_uuid","message":"Enforce nullable\u003dFalse on instance_uuid\n\nThis resolves what was requested in the TODO left by mriedem.\nThe instance_uuid must be set to nullable\u003dFalse because it\nis a core identifying variable that must always be present.\n\nChange-Id: I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a7a7711d0d15f4374c2998221ffaababa049e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a7a7711d0d15f4374c2998221ffaababa049e57"}]},"branch":"refs/heads/master"},"84f39ae7eb9b84d48182935c02460401164be6a4":{"kind":"REWORK","_number":3,"created":"2026-09-16 17:19:07.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/81/1005881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1005881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1005881/3"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:28:05.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 17:18:33.000000000","tz":-180},"subject":"Enforce nullable\u003dFalse on instance_uuid","message":"Enforce nullable\u003dFalse on instance_uuid\n\nThis resolves what was requested in the TODO left by mriedem.\nThe instance_uuid must be set to nullable\u003dFalse because it\nis a core identifying variable that must always be present.\n\nChange-Id: I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84f39ae7eb9b84d48182935c02460401164be6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84f39ae7eb9b84d48182935c02460401164be6a4"}]},"branch":"refs/heads/master"},"4e04a801343711e3a5ad624b6af14d6e2a7bfd04":{"kind":"REWORK","_number":4,"created":"2026-09-16 21:06:03.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/81/1005881/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1005881/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1005881/4"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:28:05.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 21:04:32.000000000","tz":-180},"subject":"Enforce nullable\u003dFalse on instance_uuid","message":"Enforce nullable\u003dFalse on instance_uuid\n\nThis resolves what was requested in the TODO left by mriedem.\nThe instance_uuid must be set to nullable\u003dFalse because it\nis a core identifying variable that must always be present.\n\nAssisted-By: gemini-pro-3.1\nChange-Id: I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e04a801343711e3a5ad624b6af14d6e2a7bfd04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e04a801343711e3a5ad624b6af14d6e2a7bfd04"}]},"branch":"refs/heads/master"},"7236aeddabcda093ec114b917d51c17d73cb9a77":{"kind":"REWORK","_number":5,"created":"2026-09-16 21:13:36.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/81/1005881/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1005881/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1005881/5"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:28:05.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 21:13:22.000000000","tz":-180},"subject":"Enforce nullable\u003dFalse on instance_uuid","message":"Enforce nullable\u003dFalse on instance_uuid\n\nThis resolves what was requested in the TODO left by mriedem.\nThe instance_uuid must be set to nullable\u003dFalse because it\nis a core identifying variable that must always be present.\n\nAssisted-By: gemini-pro-3.1\nChange-Id: I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7236aeddabcda093ec114b917d51c17d73cb9a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7236aeddabcda093ec114b917d51c17d73cb9a77"}]},"branch":"refs/heads/master"},"f29e0c2c318105367cd4ad596a48f73293ea733e":{"kind":"REWORK","_number":6,"created":"2026-09-17 22:01:47.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/81/1005881/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/1005881/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/1005881/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/1005881/6"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-16 13:28:05.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-09-17 22:01:31.000000000","tz":-180},"subject":"Enforce nullable\u003dFalse on instance_uuid","message":"Enforce nullable\u003dFalse on instance_uuid\n\nThis resolves what was requested in the TODO left by mriedem.\nThe instance_uuid must be set to nullable\u003dFalse because it\nis a core identifying variable that must always be present.\n\nAssisted-By: gemini-pro-3.1\nChange-Id: I9928f2d97f6b36ae28fb5dfaea39de4a3d23c254\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f29e0c2c318105367cd4ad596a48f73293ea733e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f29e0c2c318105367cd4ad596a48f73293ea733e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
