)]}'
{"id":"openstack%2Fnova~1006163","triplet_id":"openstack%2Fnova~master~I6042721db0c38d0f9d622635f0817cd6f19e16d9","project":"openstack/nova","branch":"master","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-18 10:33:33.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6042721db0c38d0f9d622635f0817cd6f19e16d9","subject":"Add missing resize.error notification","status":"NEW","created":"2026-09-18 08:13:19.000000000","updated":"2026-09-18 10:33:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7128abc341de2d68f730b050a76a9b3aa1ba0e44","_number":1006163,"virtual_id_number":1006163,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 10:33:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 08:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 09:09:33.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-18 10:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4bd15de51e5292d2aead6accea1d5cf956fa10d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-18 08:13:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"481445a45c30eaea93f634adc2edf109b114e86f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 08:51:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29d0160f16445e190ac3f6879fdcfd1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ea11d4432784fc69f8ba7c520cb44ff : SUCCESS in 20m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/608c840451924d6ebac9bb15edf30fe7 : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/21ca5546f9dc416db50be9b274186eb8 : SUCCESS in 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2224d143ac8cde9dac8a0528d3197a617344475d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 09:09:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6042721db0c38d0f9d622635f0817cd6f19e16d9 .","accounts_in_message":[],"_revision_number":1},{"id":"7128abc341de2d68f730b050a76a9b3aa1ba0e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:33:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/080c37792b534911aed8dbbece715133\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e87e7bcb4c84a42b40b2c0da4975466 : SUCCESS in 40m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3ee3e91342b344ad9336dcde34c0bdfa : SUCCESS in 1h 38m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/164c99132207451e9ac6336a70acf021 : SUCCESS in 1h 07m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81aa4011fb7e4c5e9aeacd88483fa1ef : TIMED_OUT in 1h 42m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b9ebf6729c4315830bddd6be99eece : SUCCESS in 11m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d59698ad80ba4a098517ceaaead3a9c7 : SUCCESS in 13m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/876cb1b111224f67ae8e574ab47940ad : SUCCESS in 13m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11abeab6c3d640cea4add330e718b4ad : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a552725837949eb92f515e6cf3f8e43 : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81328f2f18894539865f95993efb3782 : SUCCESS in 4m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a3ff5ffc266947b18213011a81f70d46 : SUCCESS in 1h 08m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e50cfbb201e34ea49813ca43c7596ad9 : SUCCESS in 43m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ccbf6b2be22f4eb0949a804e1f8d472a : SUCCESS in 47m 33s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/453f51f15a314780a9064e7e609619eb : SUCCESS in 40m 03s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f475b20c37e34bacbc5aa0a40b4a1390 : SUCCESS in 1h 02m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/07d638c1174b4d07ba5674ce47c535dd : SUCCESS in 58m 18s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dc5394c7fe7f489db3ad80e75e6a855d : FAILURE in 1h 27m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/fac73ea222064aaaa1513fdd70024733 : SUCCESS in 2h 11m 41s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/dfe57c5f127c44948b0de7746884d94b : SUCCESS in 52m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c66782b38263463b8159d96a7bfb351f : SUCCESS in 4m 49s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c08ef908688b424082a0bb3713c9b3b9 : SUCCESS in 30m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/177c748aa86a42139b7e661260b87dd1 : SUCCESS in 27m 26s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a61bbf2fad67445eab8703be1a4c5779 : SUCCESS in 32m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d278ea6b9cd4b43b1e4800b6528270d : FAILURE in 1h 32m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6b5a11c810a468195ddfafdba518549 : SUCCESS in 59m 56s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e3e41d32fa2343aa8eb8e4318083a60b : SUCCESS in 47m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b6f7416f563e45d1bbac85f5cbf1182a : SUCCESS in 37m 10s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/65855e53c5924c58b63ddcaa803d9425 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3fa38f9fefb7c188d398e1c3dda155f4826ba6c","revisions":{"a3fa38f9fefb7c188d398e1c3dda155f4826ba6c":{"kind":"REWORK","_number":1,"created":"2026-09-18 08:13:19.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/63/1006163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/1006163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/1006163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/1006163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/1006163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/1006163/1"}}},"commit":{"parents":[{"commit":"94de0576ca02e0668233bedfbe7e18b87316c708","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94de0576ca02e0668233bedfbe7e18b87316c708"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-17 14:29:16.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-17 14:29:24.000000000","tz":120},"subject":"Add missing resize.error notification","message":"Add missing resize.error notification\n\nIf migrate_disk_and_power_off fails with an exception during resize then\nneither instance.resize.end nor instance.resize.error emitted after the\ncorrectly emitted instance.resize.start. This patch adds the missing\nnotification.\n\nTODO:\n * test coverage\n\nChange-Id: I6042721db0c38d0f9d622635f0817cd6f19e16d9\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3fa38f9fefb7c188d398e1c3dda155f4826ba6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3fa38f9fefb7c188d398e1c3dda155f4826ba6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
