)]}'
{"id":"openstack%2Fnova~1007435","triplet_id":"openstack%2Fnova~stable%2F2025.1~I13837d1b3cc09519ac95586553666b69854c200c","project":"openstack/nova","branch":"stable/2025.1","topic":"bug/2166786","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-25 13:09:25.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I13837d1b3cc09519ac95586553666b69854c200c","subject":"Reproduce bug 2166786","status":"NEW","created":"2026-09-25 08:14:59.000000000","updated":"2026-09-25 15:26:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3fe957460d802160f2cd8912d23bc5422eca4ce7","_number":1007435,"virtual_id_number":1007435,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-25 13:09:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 08:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 13:09:25.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-25 14:31:13.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"bd4b0c78e3a5da5a6a491330da2d57930072afb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-25 08:14:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97241cbd7b5d97be594d5e25dd7a46dae22205d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:56:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ae10caca86044cd8b27e7c227c1ae40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c6ff6dbc76e4dc39170641442f6d24f : SUCCESS in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c64b19b41c4f35b6b5960dc4772aad : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2fefe0f2867426eacc1ba2d864e6971 : SUCCESS in 9m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68159bc30e3946cbaf32e4945cc4b95c : SUCCESS in 14m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5eebb16e6344e59a3c96a1b4c64ce13 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b24b6ae6fa5848d991db2e6580507e8c : SUCCESS in 8m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/65817ca5bc7c4e44ac98b5602add4405 : FAILURE in 4m 28s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bbe17fa08c9c4cda843fdcdf0f8cdccc : FAILURE in 28m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c34026834e241d297739be19d75159e : FAILURE in 26m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3d84a2ab51a7491792e228141a612b02 : FAILURE in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"aac1f957df0e39be1daba3b59d2c700cbf743b74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-25 12:33:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9af4f49bd9248596f0b503bac8f5662bd7a5fea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 13:09:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8960c2a71ca40a482312b0ec9ce4acb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36850722a3bb4104b2fc96c8a580f211 : SUCCESS in 23m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d8ed8b9a0a420090fdd2809aa5f6ba : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eafb654ea9e94590969530508565be6f : SUCCESS in 20m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3419f0c127c4929963983c339e81368 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f47eb76ca246818026b164e8e9dfc0 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31337f06e7ad497580b86f919540b6e0 : SUCCESS in 9m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8548ad87e8fd461f96b50adfb7b2abb0 : FAILURE in 5m 33s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/84f059ef35e543febcce2ca416869576 : SUCCESS in 29m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ee1bdf118cc41e7a2a63cc1db259b41 : SUCCESS in 27m 40s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e620b47f098482b938722596f77dac6 : SUCCESS in 26m 31s","accounts_in_message":[],"_revision_number":2},{"id":"193a13b1b2f5877dc884f1adbe75c8d38bfc494f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-25 13:18:40.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"262b95180a4e19e747c9c5a54ee2f9bfc0538edd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 14:31:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":1},{"id":"3fe957460d802160f2cd8912d23bc5422eca4ce7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 15:26:58.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I13837d1b3cc09519ac95586553666b69854c200c .","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c2db150bc1ee2aa20b3c124bcc86c039a52afd1","revisions":{"ec895dc48c2772ee44f3fabc814d8559abcdac23":{"kind":"REWORK","_number":1,"created":"2026-09-25 08:14:59.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/1007435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/1007435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/1007435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/1007435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/1007435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/1007435/1"}}},"commit":{"parents":[{"commit":"ce37978276744e92d91251210a1d9c2784eea375","subject":"Merge \"Fix mutating global config in websocket proxy\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce37978276744e92d91251210a1d9c2784eea375"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-25 08:14:43.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nNote that the PCI allocation in Placement (if enabled) are not affected\nby the bug.\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n(cherry picked from commit fa7cf515b1c6c608617a887000192948e314f9a4)\n(cherry picked from commit 1a26eba2222e0797352df42d338baf1bd56478f8)\n(cherry picked from commit 8e37ebe6e397dfa3766aadef228ef168b3482d90)\n(cherry picked from commit 673574a6b64e85c904147810a722073dcadad50e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec895dc48c2772ee44f3fabc814d8559abcdac23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec895dc48c2772ee44f3fabc814d8559abcdac23"}]},"branch":"refs/heads/stable/2025.1"},"2c2db150bc1ee2aa20b3c124bcc86c039a52afd1":{"kind":"REWORK","_number":2,"created":"2026-09-25 12:33:53.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/1007435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/1007435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/1007435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/1007435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/1007435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/1007435/2"}}},"commit":{"parents":[{"commit":"ce37978276744e92d91251210a1d9c2784eea375","subject":"Merge \"Fix mutating global config in websocket proxy\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce37978276744e92d91251210a1d9c2784eea375"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-08 15:06:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-09-25 12:32:43.000000000","tz":120},"subject":"Reproduce bug 2166786","message":"Reproduce bug 2166786\n\nThis reproduction shows that a cold migration that fails to access the\ndestination compute via ssh leaks PCI allocations on the destination.\n\nNote that the PCI allocation in Placement (if enabled) are not affected\nby the bug.\n\n2025.2 only change:\n  The test helper self._migrate_server need to be replaced as\n  I2790cd274a4776ab306b39df1e591e8304b63f96 is missing:\n\n  -            self._migrate_server(server, expected_state\u003d\"ACTIVE\")\n  +            self.api.post_server_action(server[\u0027id\u0027], {\u0027migrate\u0027: None})\n  +            self._wait_for_state_change(server, \u0027ACTIVE\u0027)\n\nRelated-Bug: #2166786\nChange-Id: I13837d1b3cc09519ac95586553666b69854c200c\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n(cherry picked from commit fa7cf515b1c6c608617a887000192948e314f9a4)\n(cherry picked from commit 1a26eba2222e0797352df42d338baf1bd56478f8)\n(cherry picked from commit 8e37ebe6e397dfa3766aadef228ef168b3482d90)\n(cherry picked from commit 2ea5ccb3f63228780dbfd34ae081b116021cd01a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c2db150bc1ee2aa20b3c124bcc86c039a52afd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c2db150bc1ee2aa20b3c124bcc86c039a52afd1"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
