)]}'
{"id":"openstack%2Fnova~114188","triplet_id":"openstack%2Fnova~master~I90e4153a80c859a928630809aef31e7313f08906","project":"openstack/nova","branch":"master","topic":"retrospective-revert-veto-policy","hashtags":[],"change_id":"I90e4153a80c859a928630809aef31e7313f08906","subject":"Add a policy for handling retrospective vetos","status":"MERGED","created":"2014-08-14 10:12:00.000000000","updated":"2014-09-17 00:00:01.000000000","submitted":"2014-09-16 23:59:58.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c4c3933788d0e64e5781e0206fcce82be973ba0","_number":114188,"virtual_id_number":114188,"owner":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":2,"date":"2014-09-16 23:59:58.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},{"value":1,"date":"2014-09-04 17:39:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2014-09-02 20:06:35.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-02 22:32:42.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"date":"2014-09-02 19:12:53.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"date":"2014-09-16 11:54:15.000000000","_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":2,"date":"2014-09-16 15:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":2,"date":"2014-09-03 02:32:48.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":2,"date":"2014-09-16 15:38:55.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2014-09-03 11:35:47.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":2,"date":"2014-09-09 02:06:57.000000000","_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":2,"date":"2014-09-08 22:55:36.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":2,"date":"2014-09-02 22:54:22.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":2,"date":"2014-09-16 20:02:21.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-09-16 12:29:13.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":1,"date":"2014-09-16 20:02:21.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-08-14 11:01:51.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"},{"updated":"2014-08-14 14:22:51.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2014-08-14 18:49:13.000000000","updated_by":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"reviewer":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"state":"REVIEWER"},{"updated":"2014-08-15 02:30:55.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2014-08-19 04:35:50.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-08-22 20:47:37.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2014-09-02 20:06:35.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-02 22:32:42.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-02 22:54:22.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2014-09-03 02:32:48.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2014-09-04 17:39:28.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-09-08 22:55:36.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2014-09-09 02:06:57.000000000","updated_by":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"reviewer":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"state":"REVIEWER"},{"updated":"2014-09-16 12:29:13.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2014-09-16 15:38:55.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2014-09-16 15:51:35.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2014-09-16 20:02:21.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2014-09-16 23:59:58.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"0f771345f5f470a9b94d1f212960cd8a1fdeada9","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-08-14 10:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f58c3ffa85586351d5331407ab58214a651dd940","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-14 10:12:07.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"a3a31ecb90d4b3c0ef8fc22a75d9166414c0dc29","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-08-14 10:12:21.000000000","message":"Patch Set 1:\n\nSee also this thread: http://lists.openstack.org/pipermail/openstack-dev/2014-August/thread.html#42728","accounts_in_message":[],"_revision_number":1},{"id":"02d1cb165fd338bfc0d8b47b5d11dfb2a20cdac8","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-08-14 11:01:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nSmall nit on the first paragraph that outlines what might cause such a patch to land and be missed.\n\nOtherwise +2","accounts_in_message":[],"_revision_number":1},{"id":"c2e2e630293c5eac947075a3ad7db587bad41848","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-14 11:23:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/88/114188/1/check/gate-nova-pep8/a034623 : SUCCESS in 9m 08s\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/1/check/gate-nova-docs/0ecfe9e/doc/build/html/ : SUCCESS in 18m 58s\n- gate-nova-python26 http://logs.openstack.org/88/114188/1/check/gate-nova-python26/245b03c : SUCCESS in 24m 13s\n- gate-nova-python27 http://logs.openstack.org/88/114188/1/check/gate-nova-python27/5e935c5 : SUCCESS in 20m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/88/114188/1/check/check-tempest-dsvm-full/f2a2d1b : SUCCESS in 42m 49s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/1/check/check-tempest-dsvm-postgres-full/7aaf315 : SUCCESS in 53m 50s\n- check-tempest-dsvm-neutron http://logs.openstack.org/88/114188/1/check/check-tempest-dsvm-neutron/c5d236b : SUCCESS in 50m 07s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/1/check/check-tempest-dsvm-neutron-full/21a6d97 : SUCCESS in 1h 04m 33s (non-voting)\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/1/check/check-tempest-dsvm-neutron-heat-slow/73df5aa : SUCCESS in 19m 59s\n- check-grenade-dsvm http://logs.openstack.org/88/114188/1/check/check-grenade-dsvm/980db07 : SUCCESS in 46m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/1/check/gate-tempest-dsvm-large-ops/a8a3a8c : SUCCESS in 25m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/1/check/gate-tempest-dsvm-neutron-large-ops/4371e49 : SUCCESS in 27m 23s\n- check-devstack-dsvm-cells http://logs.openstack.org/88/114188/1/check/check-devstack-dsvm-cells/f6887be : SUCCESS in 15m 10s\n- gate-nova-pylint http://logs.openstack.org/88/114188/1/check/gate-nova-pylint/6bb90b6 : SUCCESS in 24m 06s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/1/check/check-grenade-dsvm-partial-ncpu/85b98d9 : SUCCESS in 1h 00m 37s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/88/114188/1/check/check-tempest-dsvm-virtual-ironic-nv/ca6108f : SUCCESS in 26m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a55bf928a83cbf1665e142d395d21d9d69bf8aba","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-08-14 12:03:45.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1e49db9b02514f0e995f5cd6638bb874/131007_devstack_export.log : SUCCESS in 17m 05s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1bd5a63eebd848378a12dcb91fa713b0/user_001.log : SUCCESS in 23m 24s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7a7ff261433d445bbd1cbe456dedf741/131007_devstack_export.log : SUCCESS in 14m 28s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/1/check/real-db-upgrade_nova_percona_user_001:th-percona/920bf209d182414a8103a9aa08a147c4/user_001.log : SUCCESS in 23m 58s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3ff19f51eb464629a3c09f876bcc68f9/user_002.log : SUCCESS in 17m 06s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/1/check/real-db-upgrade_nova_percona_user_002:th-percona/7dce1f3e4e7043c29ff4808afb9e2fb0/user_002.log : SUCCESS in 20m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":1},{"id":"962570d9d144c3e4258f53b074deaa00552121b3","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2014-08-14 12:26:18.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks clear and sensible to me, and was discussed at http://lists.openstack.org/pipermail/openstack-dev/2014-August/042728.html","accounts_in_message":[],"_revision_number":1},{"id":"b66b043de80608fb193c8dccfeaca339f0fbaf17","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-08-14 13:22:09.000000000","message":"Patch Set 1: Code-Review+2\n\nseems fine to me","accounts_in_message":[],"_revision_number":1},{"id":"8b11d19e17f4d2226a6e0312ff7f7d3b10e5baab","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-08-14 13:50:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"986daf045769fdff216008a85b83f827476f23c0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-08-14 14:22:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nJust a +1 from me, since I agree with Nikola\u0027s statement. Change that and I vote +2.\n\nBest,\n\n-jay","accounts_in_message":[],"_revision_number":1},{"id":"649fd15979d7c095af99ff20c42a02010e786dec","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-14 18:04:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/114188/1/ : Looks good to me in 1h 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"998c74cc48c01cc688bc696dec438208b865da3b","author":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"date":"2014-08-14 18:49:13.000000000","message":"Patch Set 1: Code-Review+1\n\nI like the direction this is going in, though similar to Jay, I\u0027d like Nikola\u0027s comment addressed before merging.","accounts_in_message":[],"_revision_number":1},{"id":"6a693726e83ece2a1439067e84a047089a330328","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2014-08-15 02:30:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"389f816b9a1a519f90ebc3e7a85bb2f9b0d3fa4c","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-08-19 03:38:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"60c32bfbc3f1f9b7883f5890a31c4401bc04d5bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-19 03:38:47.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"2f32c0d8dd96143b1df688e7e71bfdad772b0137","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-08-19 03:38:54.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nI have attempted to address review concerns here to keep the ball rolling.","accounts_in_message":[],"_revision_number":1},{"id":"3ac792eb124c5c94d9e9a2e40c3079d3ffc3eab0","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-08-19 04:04:30.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e7472ed492ff454b9a0df2a75894e35b/131007_devstack_export.log : SUCCESS in 14m 09s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c66d3a82fc024bf0a2e0a422f57b30fe/user_001.log : SUCCESS in 24m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9df44ab869f6481796b233f8278ac989/131007_devstack_export.log : SUCCESS in 15m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/2/check/real-db-upgrade_nova_percona_user_001:th-percona/d783469edf854458946cd8f6ded8d8fc/user_001.log : SUCCESS in 24m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/fb505a9f8a7f4c93bee09467c7628a20/user_002.log : SUCCESS in 20m 11s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/2/check/real-db-upgrade_nova_percona_user_002:th-percona/ae615ea594ac445a98b7dda2c9b06ae0/user_002.log : SUCCESS in 20m 27s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"e1c0b2862bd85bdbf8abf32ee4dce32e86d6050d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-19 04:33:20.000000000","message":"Patch Set 2:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/114188/2/ : Test run failed. in 54m 08s","accounts_in_message":[],"_revision_number":2},{"id":"27343365d743053d72fe90502a6f0cb1fb98c0ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-19 04:35:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/88/114188/2/check/gate-nova-pep8/95f414a : SUCCESS in 9m 29s\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/2/check/gate-nova-docs/bdf6f39/doc/build/html/ : SUCCESS in 12m 58s\n- gate-nova-python26 http://logs.openstack.org/88/114188/2/check/gate-nova-python26/998d30c : SUCCESS in 17m 23s\n- gate-nova-python27 http://logs.openstack.org/88/114188/2/check/gate-nova-python27/f728678 : SUCCESS in 22m 34s\n- check-tempest-dsvm-full http://logs.openstack.org/88/114188/2/check/check-tempest-dsvm-full/c648ecf : SUCCESS in 40m 22s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/2/check/check-tempest-dsvm-postgres-full/86e19bc : SUCCESS in 38m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/2/check/check-tempest-dsvm-neutron-full/0a0ef88 : SUCCESS in 56m 10s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/2/check/check-tempest-dsvm-neutron-heat-slow/29aac74 : SUCCESS in 25m 00s\n- check-grenade-dsvm http://logs.openstack.org/88/114188/2/check/check-grenade-dsvm/83676e3 : FAILURE in 30m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/2/check/gate-tempest-dsvm-large-ops/d46b5ab : SUCCESS in 25m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/2/check/gate-tempest-dsvm-neutron-large-ops/96ea4c6 : SUCCESS in 23m 10s\n- check-devstack-dsvm-cells http://logs.openstack.org/88/114188/2/check/check-devstack-dsvm-cells/f6ca909 : SUCCESS in 17m 24s\n- gate-nova-pylint http://logs.openstack.org/88/114188/2/check/gate-nova-pylint/35bb4bd : SUCCESS in 22m 08s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/2/check/check-grenade-dsvm-partial-ncpu/a946b24 : SUCCESS in 43m 28s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/88/114188/2/check/check-tempest-dsvm-virtual-ironic-nv/1b746f9 : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c07f049880cebabc6eaf225824d8e0edecc618e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-08-19 04:35:50.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1331274\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"0a2c305f13cf9a2b52b5ad81783fe4577b614471","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-19 05:03:44.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/88/114188/2/22129/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"064189c43b787e1d5ad1a5e75a543a48fe17ef46","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2014-08-19 11:07:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3036c4ff0c5e5f0409d5d20756971dab90c33872","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-08-20 15:23:58.000000000","message":"Patch Set 2:\n\n(1 comment)\n\noverall looks good, but not sure if we should explicitly call out the specs case. And maybe a warning about how it looks if we revert things a lot.","accounts_in_message":[],"_revision_number":2},{"id":"2146e3291dc40f4ccc62b3b2a93350f4820b8704","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-08-20 16:04:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThis seems good to me, but not approving is it feels like we should get a wider agreement, given joe\u0027s comment.","accounts_in_message":[],"_revision_number":2},{"id":"f2b1778701a88db2c884e4ac99a394cb2e693db1","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-08-22 14:54:49.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI\u0027m cool with Michael\u0027s changes, thanks Michael\n\nSmall issue with the dangling link inline, so I\u0027ll push another tweak","accounts_in_message":[],"_revision_number":2},{"id":"568cec3f369e4cc816d75622cd924de8064c5c3b","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-08-22 14:54:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"27f8f8b83deafed74f8ff83898474b2efd2ec117","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-08-22 15:27:53.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/67fb93bbd5c74406b7e350dcfbe65c53/131007_devstack_export.log : SUCCESS in 13m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a1ca3b2d50664830bfc38b803001563e/user_001.log : SUCCESS in 21m 32s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d7ea05c21e04453e98c69808cf6ef113/131007_devstack_export.log : SUCCESS in 14m 03s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/3/check/real-db-upgrade_nova_percona_user_001:th-percona/b757007cd46b4277b6a5b478037a123b/user_001.log : SUCCESS in 24m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ac22eb5a8d4a4af7bc7dc8edf8eeb6da/user_002.log : SUCCESS in 16m 53s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/3/check/real-db-upgrade_nova_percona_user_002:th-percona/298c7cdd884845e089985f143f2d400b/user_002.log : SUCCESS in 20m 55s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"9c7fd979c845760c6435100671d76b2c15771657","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-08-22 15:43:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"60c86208e018b9aee74c74dbdcc62a5e72175a6c","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2014-08-22 16:25:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8038d1cf96f4c682f9a2a7fea2db8d7cd28bec5b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-22 16:56:35.000000000","message":"Patch Set 3: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/88/114188/3/22897/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"4579d3de7c4826df31fc843ba63ead079b8015d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-22 16:59:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/88/114188/3/check/gate-nova-pep8/b720e80 : SUCCESS in 9m 28s\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/3/check/gate-nova-docs/72b1d0a/doc/build/html/ : SUCCESS in 12m 22s\n- gate-nova-python26 http://logs.openstack.org/88/114188/3/check/gate-nova-python26/456c05c : SUCCESS in 25m 52s\n- gate-nova-python27 http://logs.openstack.org/88/114188/3/check/gate-nova-python27/8e7427c : SUCCESS in 15m 18s\n- check-tempest-dsvm-full http://logs.openstack.org/88/114188/3/check/check-tempest-dsvm-full/9d9f875 : SUCCESS in 46m 21s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/3/check/check-tempest-dsvm-postgres-full/5f92070 : SUCCESS in 48m 32s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/3/check/check-tempest-dsvm-neutron-full/44b0ddb : SUCCESS in 48m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/3/check/check-tempest-dsvm-neutron-heat-slow/1fe3a1e : SUCCESS in 24m 46s\n- check-grenade-dsvm http://logs.openstack.org/88/114188/3/check/check-grenade-dsvm/56b2d75 : SUCCESS in 42m 56s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/3/check/gate-tempest-dsvm-large-ops/befd900 : SUCCESS in 20m 21s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/3/check/gate-tempest-dsvm-neutron-large-ops/7f21eed : SUCCESS in 26m 27s\n- check-devstack-dsvm-cells http://logs.openstack.org/88/114188/3/check/check-devstack-dsvm-cells/8bedd95 : SUCCESS in 18m 18s\n- gate-nova-pylint http://logs.openstack.org/88/114188/3/check/gate-nova-pylint/ff7b47d : SUCCESS in 17m 54s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/3/check/check-grenade-dsvm-partial-ncpu/755c9c7 : SUCCESS in 36m 07s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/88/114188/3/check/check-tempest-dsvm-virtual-ironic-nv/6c70e1a : FAILURE in 46m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34c30651aecf1278454e5b10e43d1be10827d8d6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-22 17:13:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/114188/3/ : Looks good to me in 1h 23m 49s","accounts_in_message":[],"_revision_number":3},{"id":"3196c58d530adaba6fa9269a772ce0335e430dcb","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2014-08-22 20:47:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bbf9e9d35c29e0e4da81ef0dbf184080aa6319ac","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-08-25 18:19:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"62cea14a46acacf8cf0e48ec622e08e2e7bb477d","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-08-25 21:27:12.000000000","message":"Patch Set 3:\n\nThoughts on how this should work across release boundaries for using facing changes?\n\nIf Bob adds an API change in Juno, and Carl realizes something is very bad in that API change in Kilo, can we still revert even though the change is in a stable release?","accounts_in_message":[],"_revision_number":3},{"id":"7119a7493bd896be95d8c1d4b845ba1ba8c80a25","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-08-27 12:49:23.000000000","message":"Patch Set 3: Code-Review+2\n\n\u003e If Bob adds an API change in Juno, and Carl realizes something is very bad in that API change in Kilo, can we still revert even though the change is in a stable release?\n\nIf we\u0027ve actually released something that affects API, or would affect operation of user deployments, then we can\u0027t do a plain automatic revert to the patch. Either we\u0027d have to manually write any necessary compatibility code to avoid breaking user deployments, or we\u0027d have to go through the normal deprecation cycle we have to general removal of features. ie deprecated in release N, delete in N+1.","accounts_in_message":[],"_revision_number":3},{"id":"91dbbb7240d70787772cd247da8b8a41cb5f7670","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-09-02 18:51:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d49661a534af81f03f62a5af1511bf74448ec30","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-09-02 19:12:53.000000000","message":"Patch Set 4:\n\nI\u0027ve had a go at a tiny tweak to handle this case. Checkout rule 0 in the policy for details.","accounts_in_message":[],"_revision_number":4},{"id":"8e699edc42fe1621459449523ca143b33aa67c13","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-02 20:06:35.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/88/114188/4/24603/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"84622b5b1c89f2ed43c7a3b6b9fbd4c4eeb94918","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-02 21:51:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/88/114188/4/check/gate-nova-pep8/68a008c : SUCCESS in 12m 18s\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/4/check/gate-nova-docs/592bc77/doc/build/html/ : SUCCESS in 17m 05s\n- gate-nova-python26 http://logs.openstack.org/88/114188/4/check/gate-nova-python26/7411bd8 : SUCCESS in 23m 54s\n- gate-nova-python27 http://logs.openstack.org/88/114188/4/check/gate-nova-python27/2550cba : SUCCESS in 14m 16s\n- check-tempest-dsvm-full http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-full/36a2b2e : SUCCESS in 57m 48s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-postgres-full/3e9de45 : SUCCESS in 1h 17m 19s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-neutron-full/93bf0ee : SUCCESS in 54m 20s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-neutron-heat-slow/084e0e9 : SUCCESS in 25m 06s\n- check-grenade-dsvm http://logs.openstack.org/88/114188/4/check/check-grenade-dsvm/4bfb7f8 : SUCCESS in 38m 39s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/4/check/gate-tempest-dsvm-large-ops/1263e5f : SUCCESS in 25m 10s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/4/check/gate-tempest-dsvm-neutron-large-ops/90d5828 : SUCCESS in 19m 11s\n- check-devstack-dsvm-cells http://logs.openstack.org/88/114188/4/check/check-devstack-dsvm-cells/6dc34a1 : SUCCESS in 13m 09s\n- gate-nova-pylint http://logs.openstack.org/88/114188/4/check/gate-nova-pylint/8398f02 : SUCCESS in 23m 19s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/4/check/check-grenade-dsvm-partial-ncpu/743cf15 : SUCCESS in 45m 13s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-virtual-ironic-nv/bb32254 : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a1933bef07ffd9fdfbaac5327cb8daa522a664da","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-02 22:32:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/114188/4/ : Looks good to me in 1h 34m 12s","accounts_in_message":[],"_revision_number":4},{"id":"847458d857130a8255892da5e72659b9bda97d26","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-09-02 22:54:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"62a591c1257254fdb193a45524fef23e1ce68d45","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2014-09-03 02:32:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"866c492bf06127573aec5d49716a2ccea7511cfd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-03 08:42:37.000000000","message":"Patch Set 4: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-mysql-rax-syd-1957040.slave.openstack.org-0/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9fa83666804145388a173e290fa3eb3e/user_001.log : SUCCESS in 25m 29s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1957214.slave.openstack.org-0/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 16s\n- real-db-upgrade_nova_percona_user_001:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1957094.slave.openstack.org-4/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 3m 37s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-mysql-rax-syd-1957242.slave.openstack.org-5/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 3m 37s\n- real-db-upgrade_nova_percona_user_002:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1957215.slave.openstack.org-5/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 21s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"468619c9df7a5defd64c8e3123042d37921556b3","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-09-03 11:35:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f547de02a519baf2790a11bc472cf13d54dfc42d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-04 17:39:28.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a30fda55e6754a3dbb57557c202bc189/131007_devstack_export.log : SUCCESS in 14m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d5ac87cb5321468ea96e476306c10ede/user_001.log : SUCCESS in 22m 10s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8feb9dc64a2a4af096ef7f1fba87a378/131007_devstack_export.log : SUCCESS in 16m 20s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_percona_user_001:th-percona/c8ee983ff22d4a6883ecfa8a25ced346/user_001.log : SUCCESS in 25m 53s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c734b98de344786a2dbce75c728f1e3/user_002.log : SUCCESS in 19m 02s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/88/114188/4/check/real-db-upgrade_nova_percona_user_002:th-percona/c0f5b004f84a470ea338372f45ccbbcb/user_002.log : SUCCESS in 20m 03s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"722f964e35b188a42fce0f2b7d6f99959baa2231","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-09-08 22:55:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cb017357829211e086d3cfdad08def69fb8c9400","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2014-09-09 02:06:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f466ae4e82b8abf5e988b701edd05341dd7aef7e","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-09-16 11:54:15.000000000","message":"Patch Set 4:\n\nAnyone care to approve this?","accounts_in_message":[],"_revision_number":4},{"id":"6e1bc6800e16fd1aafd686b14df62d5f79cd2146","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-09-16 12:29:13.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f97be6ecf534095f1d74b9d47ded8cc7d40cfbe4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 12:29:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"bb8f60eb267ca08240344db6c58be1ca810b64f6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 14:51:46.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/4/gate/gate-nova-docs/fce2d04/doc/build/html/ : SUCCESS in 16m 37s\n- gate-nova-pep8 http://logs.openstack.org/88/114188/4/gate/gate-nova-pep8/22099c2 : SUCCESS in 8m 58s\n- gate-nova-python26 http://logs.openstack.org/88/114188/4/gate/gate-nova-python26/a18dfb2 : FAILURE in 28s\n- gate-nova-python27 http://logs.openstack.org/88/114188/4/gate/gate-nova-python27/cd26ca9 : SUCCESS in 14m 34s\n- gate-tempest-dsvm-full http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-full/8ed18e5 : SUCCESS in 46m 18s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-postgres-full/8d3fe6e : SUCCESS in 1h 18m 38s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-neutron-full/debb8c2 : SUCCESS in 1h 06m 08s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-neutron-heat-slow/0c097c6 : SUCCESS in 19m 11s\n- gate-grenade-dsvm http://logs.openstack.org/88/114188/4/gate/gate-grenade-dsvm/ce24ae4 : SUCCESS in 44m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-large-ops/84116ff : SUCCESS in 25m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-neutron-large-ops/00c1704 : SUCCESS in 19m 23s\n- gate-devstack-dsvm-cells http://logs.openstack.org/88/114188/4/gate/gate-devstack-dsvm-cells/24cfe14 : SUCCESS in 16m 38s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/4/gate/gate-grenade-dsvm-partial-ncpu/2c78f1c : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":4},{"id":"23bbdde79870ef8e664dc19f6682bbdb8ebdba81","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-09-16 15:31:23.000000000","message":"Patch Set 4:\n\nI would actually like to see a few more +2s from nova-drivers (including the PTL on this before merging)","accounts_in_message":[],"_revision_number":4},{"id":"de02d397ef7a3b4172bdd8c2dc6477a6adf762eb","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-09-16 15:38:55.000000000","message":"Patch Set 4: Code-Review+2\n\nI like getting this unwritten rule written down. Thanks.","accounts_in_message":[],"_revision_number":4},{"id":"48dc73a1e2bc05255e9e84991d2b623617ffad15","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2014-09-16 15:51:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1f92e3cce9cbef7eb45f3535f062ac0bd52f78a8","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-09-16 20:02:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1e697dd720d1adb62f9b79097de192ab018d7bea","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-09-16 20:02:33.000000000","message":"Patch Set 4:\n\nreverify","accounts_in_message":[],"_revision_number":4},{"id":"cf3cd916a8b338e4e90cb39d06e26f44a0b9500d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-16 21:24:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- http://64.119.130.115/114188/4/ : Looks good to me in 1h 21m 37s","accounts_in_message":[],"_revision_number":4},{"id":"af223dc1e63d3dd88f2b441f1929d1ed48bdd916","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 22:13:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/88/114188/4/check/gate-nova-pep8/0bf9f87 : SUCCESS in 8m 42s\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/4/check/gate-nova-docs/742f6a4/doc/build/html/ : SUCCESS in 13m 38s\n- gate-nova-python26 http://logs.openstack.org/88/114188/4/check/gate-nova-python26/d9537bb : SUCCESS in 16m 55s\n- gate-nova-python27 http://logs.openstack.org/88/114188/4/check/gate-nova-python27/41c11cb : SUCCESS in 14m 39s\n- check-tempest-dsvm-full http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-full/592e7dc : SUCCESS in 56m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-postgres-full/8e1b451 : SUCCESS in 1h 26m 21s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-neutron-full/4cd753e : SUCCESS in 1h 06m 40s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-neutron-heat-slow/7b6796a : SUCCESS in 24m 00s\n- check-grenade-dsvm http://logs.openstack.org/88/114188/4/check/check-grenade-dsvm/65be242 : SUCCESS in 57m 09s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/4/check/gate-tempest-dsvm-large-ops/c18a9a6 : SUCCESS in 18m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/4/check/gate-tempest-dsvm-neutron-large-ops/2d1fd1e : SUCCESS in 23m 24s\n- check-devstack-dsvm-cells http://logs.openstack.org/88/114188/4/check/check-devstack-dsvm-cells/8ed847b : SUCCESS in 17m 20s\n- gate-nova-pylint http://logs.openstack.org/88/114188/4/check/gate-nova-pylint/14e1b38 : SUCCESS in 15m 32s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/4/check/check-grenade-dsvm-partial-ncpu/6b6b109 : SUCCESS in 44m 22s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/88/114188/4/check/check-tempest-dsvm-ironic-pxe_ssh-nv/d81879b : SUCCESS in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9371401813ca17a18c9f2434f3b1417e4fcecd08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 22:13:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"29fdef0f1ae036d5c4464a3980325c89ab3e07c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 23:59:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded.\n\n- gate-nova-docs http://docs-draft.openstack.org/88/114188/4/gate/gate-nova-docs/0655d7d/doc/build/html/ : SUCCESS in 16m 14s\n- gate-nova-pep8 http://logs.openstack.org/88/114188/4/gate/gate-nova-pep8/e043286 : SUCCESS in 8m 57s\n- gate-nova-python26 http://logs.openstack.org/88/114188/4/gate/gate-nova-python26/a96ee37 : SUCCESS in 15m 32s\n- gate-nova-python27 http://logs.openstack.org/88/114188/4/gate/gate-nova-python27/4e5e187 : SUCCESS in 15m 23s\n- gate-tempest-dsvm-full http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-full/d3e5bdc : SUCCESS in 44m 00s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-postgres-full/0bc42df : SUCCESS in 1h 08m 59s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-neutron-full/c3f57fc : SUCCESS in 1h 07m 52s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-neutron-heat-slow/3fe6ed1 : SUCCESS in 25m 10s\n- gate-grenade-dsvm http://logs.openstack.org/88/114188/4/gate/gate-grenade-dsvm/116cc65 : SUCCESS in 46m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-large-ops/79938b0 : SUCCESS in 24m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/88/114188/4/gate/gate-tempest-dsvm-neutron-large-ops/b121572 : SUCCESS in 17m 52s\n- gate-devstack-dsvm-cells http://logs.openstack.org/88/114188/4/gate/gate-devstack-dsvm-cells/f42a467 : SUCCESS in 13m 35s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/88/114188/4/gate/gate-grenade-dsvm-partial-ncpu/e348ae4 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":4},{"id":"ff282bc01a7878e6a03edc0bea1981cabba5afd7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-17 00:00:01.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e48da64a15f396de03464cf96fbabdc59589f72a","revisions":{"1b496c584ce3f2958499b824fa184c5b90a9ff3f":{"kind":"REWORK","_number":1,"created":"2014-08-14 10:12:00.000000000","uploader":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"ref":"refs/changes/88/114188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/114188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/114188/1"}}},"commit":{"parents":[{"commit":"0c70d91c054e3ad159a8714acb164c24df3036b9","subject":"Merge \"VMware: revert deletion of cleanup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c70d91c054e3ad159a8714acb164c24df3036b9"}]}],"author":{"name":"Mark McLoughlin","email":"markmc@redhat.com","date":"2014-08-14 10:08:52.000000000","tz":60},"committer":{"name":"Mark McLoughlin","email":"markmc@redhat.com","date":"2014-08-14 10:08:52.000000000","tz":60},"subject":"Add a policy for handling retrospective vetos","message":"Add a policy for handling retrospective vetos\n\nFor some time, we\u0027ve been informally implementing a \"revert early\"\npolicy around handling -2 worthy complaints after a patch has been\nmerged. These reverts can be contentious, so it\u0027s appropriate we make\nit very clear what the policy and procedure here is in order to lessen\nthe contentiousness of these reverts.\n\nChange-Id: I90e4153a80c859a928630809aef31e7313f08906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b496c584ce3f2958499b824fa184c5b90a9ff3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b496c584ce3f2958499b824fa184c5b90a9ff3f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0c70d91c054e3ad159a8714acb164c24df3036b9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3423f35a7059bdfd613e042b775df95063350411":{"kind":"REWORK","_number":2,"created":"2014-08-19 03:38:44.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/88/114188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/114188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/114188/2"}}},"commit":{"parents":[{"commit":"0c70d91c054e3ad159a8714acb164c24df3036b9","subject":"Merge \"VMware: revert deletion of cleanup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c70d91c054e3ad159a8714acb164c24df3036b9"}]}],"author":{"name":"Mark McLoughlin","email":"markmc@redhat.com","date":"2014-08-14 10:08:52.000000000","tz":60},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2014-08-19 03:37:44.000000000","tz":600},"subject":"Add a policy for handling retrospective vetos","message":"Add a policy for handling retrospective vetos\n\nFor some time, we\u0027ve been informally implementing a \"revert early\"\npolicy around handling -2 worthy complaints after a patch has been\nmerged. These reverts can be contentious, so it\u0027s appropriate we make\nit very clear what the policy and procedure here is in order to lessen\nthe contentiousness of these reverts.\n\nCo-authored-by: Michael Still \u003cmikal@stillhq.com\u003e\n\nChange-Id: I90e4153a80c859a928630809aef31e7313f08906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3423f35a7059bdfd613e042b775df95063350411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3423f35a7059bdfd613e042b775df95063350411"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0c70d91c054e3ad159a8714acb164c24df3036b9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"19a6721c60c2b624fa5e8a005ed12aeb85b9cd55":{"kind":"REWORK","_number":3,"created":"2014-08-22 14:54:49.000000000","uploader":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"ref":"refs/changes/88/114188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/114188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/114188/3"}}},"commit":{"parents":[{"commit":"0c70d91c054e3ad159a8714acb164c24df3036b9","subject":"Merge \"VMware: revert deletion of cleanup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c70d91c054e3ad159a8714acb164c24df3036b9"}]}],"author":{"name":"Mark McLoughlin","email":"markmc@redhat.com","date":"2014-08-14 10:08:52.000000000","tz":60},"committer":{"name":"Mark McLoughlin","email":"markmc@redhat.com","date":"2014-08-22 14:50:06.000000000","tz":60},"subject":"Add a policy for handling retrospective vetos","message":"Add a policy for handling retrospective vetos\n\nFor some time, we\u0027ve been informally implementing a \"revert early\"\npolicy around handling -2 worthy complaints after a patch has been\nmerged. These reverts can be contentious, so it\u0027s appropriate we make\nit very clear what the policy and procedure here is in order to lessen\nthe contentiousness of these reverts.\n\nCo-authored-by: Michael Still \u003cmikal@stillhq.com\u003e\n\nChange-Id: I90e4153a80c859a928630809aef31e7313f08906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/19a6721c60c2b624fa5e8a005ed12aeb85b9cd55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/19a6721c60c2b624fa5e8a005ed12aeb85b9cd55"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0c70d91c054e3ad159a8714acb164c24df3036b9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e48da64a15f396de03464cf96fbabdc59589f72a":{"kind":"REWORK","_number":4,"created":"2014-09-02 18:51:53.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/88/114188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/114188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/114188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/114188/4"}}},"commit":{"parents":[{"commit":"0c70d91c054e3ad159a8714acb164c24df3036b9","subject":"Merge \"VMware: revert deletion of cleanup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c70d91c054e3ad159a8714acb164c24df3036b9"}]}],"author":{"name":"Mark McLoughlin","email":"markmc@redhat.com","date":"2014-08-14 10:08:52.000000000","tz":60},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2014-09-02 18:51:29.000000000","tz":600},"subject":"Add a policy for handling retrospective vetos","message":"Add a policy for handling retrospective vetos\n\nFor some time, we\u0027ve been informally implementing a \"revert early\"\npolicy around handling -2 worthy complaints after a patch has been\nmerged. These reverts can be contentious, so it\u0027s appropriate we make\nit very clear what the policy and procedure here is in order to lessen\nthe contentiousness of these reverts.\n\nCo-authored-by: Michael Still \u003cmikal@stillhq.com\u003e\n\nChange-Id: I90e4153a80c859a928630809aef31e7313f08906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e48da64a15f396de03464cf96fbabdc59589f72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e48da64a15f396de03464cf96fbabdc59589f72a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0c70d91c054e3ad159a8714acb164c24df3036b9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
