)]}'
{"id":"openstack%2Fnova~118675","triplet_id":"openstack%2Fnova~master~I5cf075f38696110d2890cfa1ea488ca459d791a3","project":"openstack/nova","branch":"master","topic":"optimize_in_floating_ip_bulk_create","hashtags":[],"change_id":"I5cf075f38696110d2890cfa1ea488ca459d791a3","subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","status":"MERGED","created":"2014-09-03 16:03:58.000000000","updated":"2014-12-03 20:56:57.000000000","submitted":"2014-12-03 20:56:54.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b573eb1ab90438c05d573656ef6fbc513aabb183","_number":118675,"virtual_id_number":118675,"owner":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":2,"date":"2014-12-03 20:56:54.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2014-12-01 17:52:19.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2014-12-03 11:49:25.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":1,"date":"2014-12-01 17:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},{"value":0,"date":"2014-12-02 14:58:04.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2014-12-01 17:50:15.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-12-01 18:31:40.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"date":"2014-12-02 15:39:13.000000000","_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},{"value":0,"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},{"value":0,"date":"2014-12-03 17:53:10.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"recommended":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"all":[{"value":1,"date":"2014-12-02 12:26:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":1,"date":"2014-12-02 15:03:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":2,"date":"2014-12-02 00:22:47.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":1,"date":"2014-12-03 02:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-12-02 15:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},{"value":2,"date":"2014-12-03 11:04:35.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},{"value":0,"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"all":[{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},{"value":1,"date":"2014-12-03 11:04:35.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},{"value":0,"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},{"value":0,"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},{"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-09-05 11:02:03.000000000","updated_by":{"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},"reviewer":{"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},"state":"REVIEWER"},{"updated":"2014-10-01 11:36:00.000000000","updated_by":{"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},"reviewer":{"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},"state":"REVIEWER"},{"updated":"2014-12-01 16:59:59.000000000","updated_by":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"reviewer":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"state":"REVIEWER"},{"updated":"2014-12-01 17:32:16.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-01 17:50:15.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-01 17:52:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-01 18:31:40.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-02 00:22:47.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2014-12-02 12:26:09.000000000","updated_by":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"reviewer":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"state":"REVIEWER"},{"updated":"2014-12-02 14:58:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-02 15:03:10.000000000","updated_by":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"reviewer":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"state":"REVIEWER"},{"updated":"2014-12-02 15:08:56.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2014-12-03 02:15:38.000000000","updated_by":{"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},"reviewer":{"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},"state":"REVIEWER"},{"updated":"2014-12-03 11:04:35.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2014-12-03 11:49:25.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-12-03 17:53:10.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2014-12-03 20:56:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"cdd0a48c250a5a9dc0616771427985dc92a4275d","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-03 16:03:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e92c2366c50a3522ab3c7bd477e556dba954cb4a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-03 19:37:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/1/ : Looks good to me in 1h 25m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1de02f29196e463f1f637d118a4e3c107f29b2fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-03 19:38:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/1/check/gate-nova-pep8/ca0ab7a : SUCCESS in 12m 12s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/1/check/gate-nova-docs/dc9d4ec/doc/build/html/ : SUCCESS in 23m 17s\n- gate-nova-python26 http://logs.openstack.org/75/118675/1/check/gate-nova-python26/784b882 : FAILURE in 17m 37s\n- gate-nova-python27 http://logs.openstack.org/75/118675/1/check/gate-nova-python27/f548101 : FAILURE in 21m 56s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/1/check/check-tempest-dsvm-full/003e60d : SUCCESS in 1h 02m 01s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/1/check/check-tempest-dsvm-postgres-full/85b11d1 : SUCCESS in 1h 04m 59s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/1/check/check-tempest-dsvm-neutron-full/cc509f5 : SUCCESS in 52m 52s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/1/check/check-tempest-dsvm-neutron-heat-slow/ecfee62 : SUCCESS in 19m 46s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/1/check/check-grenade-dsvm/cae0cda : SUCCESS in 38m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/1/check/gate-tempest-dsvm-large-ops/c52a203 : SUCCESS in 19m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/1/check/gate-tempest-dsvm-neutron-large-ops/cb90413 : SUCCESS in 24m 01s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/1/check/check-devstack-dsvm-cells/7bd6784 : SUCCESS in 15m 02s\n- gate-nova-pylint http://logs.openstack.org/75/118675/1/check/gate-nova-pylint/7f52ff2 : FAILURE in 16m 13s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/1/check/check-grenade-dsvm-partial-ncpu/1601ccf : SUCCESS in 40m 33s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/75/118675/1/check/check-tempest-dsvm-virtual-ironic-nv/52be61a : SUCCESS in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d20bea3a0504f8b2bf405eebc1f3f579aef3dc2d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-03 19:42:51.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/1/24782/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"dd9b3ca69e358b3e7e419dafd235892e52a9734c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-03 22:03:47.000000000","message":"Patch Set 1: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-mysql-rax-syd-1957153.slave.openstack.org-0/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 17s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-mysql-rax-syd-1957151.slave.openstack.org-4/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 50s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1956749.slave.openstack.org-0/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 05s\n- real-db-upgrade_nova_percona_user_001:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1957271.slave.openstack.org-4/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 3m 38s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-mysql-rax-syd-1957176.slave.openstack.org-5/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1957296.slave.openstack.org-5/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 23s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6b28bf228339ff5138b154dad01a43ee9b0a2f3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-04 08:24:22.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a3e7f5fa5e904bcfa45c55ccd628b4cf/131007_devstack_export.log : SUCCESS in 15m 33s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2412eae7f7604062acfbc7d2a1372acd/user_001.log : SUCCESS in 22m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1066a210d05142e3bfc5b761919c99e6/131007_devstack_export.log : SUCCESS in 15m 20s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/1/check/real-db-upgrade_nova_percona_user_001:th-percona/90fa399645d04837871f5b7468f73ab4/user_001.log : SUCCESS in 20m 59s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/df9a8d2358c14aaea7918cfc9ea69dc8/user_002.log : SUCCESS in 22m 09s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/1/check/real-db-upgrade_nova_percona_user_002:th-percona/5fbaa44070c14cac99137021f7e9f1f4/user_002.log : SUCCESS in 20m 44s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"06e5247c0743beaf3469d3b07e6f0b4c54dd7bdd","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-04 15:54:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d00701ac5bd3608a4e025eee3779a2ddf7c0c925","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-04 16:46:44.000000000","message":"Patch Set 2:\n\nFirst patch was incorrect","accounts_in_message":[],"_revision_number":2},{"id":"8636b963b101e85ec48764b412b557cf529d7818","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-04 16:46:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c8819c09e93f347bacdb06d383d2e2f6cf265673","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-04 17:09:47.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/2/24926/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"ff703ba424b57cbb78ff644391f8a168b993552c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-04 18:55:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/2/ : Looks good to me in 1h 23m 11s","accounts_in_message":[],"_revision_number":2},{"id":"39c6faf72fcedc3802da2032e38d2cf501395ad1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-04 21:05:41.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2401c0de74814be292e5f6de79e39349/131007_devstack_export.log : SUCCESS in 14m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/335d755afdcf40d39dff6f5a19b8f1f4/user_001.log : SUCCESS in 20m 49s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d8cbdd76135246fba6dc8f5566e65cac/131007_devstack_export.log : SUCCESS in 17m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/2/check/real-db-upgrade_nova_percona_user_001:th-percona/94cd087473e740ba8830435f74c3372e/user_001.log : SUCCESS in 27m 22s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5c544fdeed2d4acab6bcd28978cad42a/user_002.log : SUCCESS in 20m 15s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/2/check/real-db-upgrade_nova_percona_user_002:th-percona/96b43b158d6a490a8f003dd2cfbe60a9/user_002.log : SUCCESS in 18m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"12069d6a1a449b0ecbf6ecffb59a66f3ade9c0b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-04 22:27:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/2/check/gate-nova-pep8/e2ad19a : SUCCESS in 13m 40s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/2/check/gate-nova-docs/4ef8bb1/doc/build/html/ : SUCCESS in 25m 45s\n- gate-nova-python26 http://logs.openstack.org/75/118675/2/check/gate-nova-python26/aa7683f : FAILURE in 16m 24s\n- gate-nova-python27 http://logs.openstack.org/75/118675/2/check/gate-nova-python27/165aa2a : FAILURE in 20m 49s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/2/check/check-tempest-dsvm-full/9ad6ac1 : SUCCESS in 48m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/2/check/check-tempest-dsvm-postgres-full/5204fa5 : SUCCESS in 59m 43s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/2/check/check-tempest-dsvm-neutron-full/d2d5229 : SUCCESS in 1h 07m 42s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/2/check/check-tempest-dsvm-neutron-heat-slow/0520f10 : SUCCESS in 19m 59s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/2/check/check-grenade-dsvm/83ebd57 : SUCCESS in 47m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/2/check/gate-tempest-dsvm-large-ops/19ba338 : SUCCESS in 19m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/2/check/gate-tempest-dsvm-neutron-large-ops/6ab8beb : SUCCESS in 20m 20s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/2/check/check-devstack-dsvm-cells/00510d0 : SUCCESS in 14m 21s\n- gate-nova-pylint http://logs.openstack.org/75/118675/2/check/gate-nova-pylint/74aa559 : FAILURE in 14m 30s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/2/check/check-grenade-dsvm-partial-ncpu/6fa856a : FAILURE in 43m 05s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/75/118675/2/check/check-tempest-dsvm-virtual-ironic-nv/e0a3f68 : SUCCESS in 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd56d8ae890c14dc0e87c8234389104d7740acc4","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-09-04 22:49:04.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-partial-ncpu: https://bugs.launchpad.net/bugs/1363552 https://bugs.launchpad.net/bugs/1349617\n- gate-nova-python26: unrecognized error\n- gate-nova-python27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"dad9e0ba82fa75022caf3b559fd7b72050dc2977","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-05 09:14:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"965be515319d6b3d3b90c2a3baeff0a2416a9a75","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 09:34:36.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"a24028f06b56a4b00e4a7235db9c12dfb391c5c4","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-05 09:37:50.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/268f20aa8fdf465ba9031a6e31887a88/131007_devstack_export.log : SUCCESS in 17m 22s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ea2eb135f44147d4920eb4aef8855f35/user_001.log : SUCCESS in 22m 30s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a62d8534bbc0442f99841a23f4f983bb/131007_devstack_export.log : SUCCESS in 15m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/3/check/real-db-upgrade_nova_percona_user_001:th-percona/d131a991082c47be87ba9f6db0c3b6f8/user_001.log : SUCCESS in 22m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c5215dcf0f00453db60a73af177ff34c/user_002.log : SUCCESS in 19m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/3/check/real-db-upgrade_nova_percona_user_002:th-percona/e6f20558cd464880b2905e7474fa13dc/user_002.log : SUCCESS in 19m 58s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"c577e8424f7b69fdcb680db474d6e7a8f467695b","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-05 10:02:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9a87fc2467adc66c210802b2d43ce6aa46443a5f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 10:03:48.000000000","message":"Patch Set 4:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"1c6fc9fd3d85ea30b51e774351b6cf8bf3bb19aa","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-05 10:19:38.000000000","message":"Patch Set 4: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/4/25031/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"a2d9966e49d91132f02cdd48c75b642cca083e79","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-05 10:29:03.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/341bdab63444468388d2dc96dac7e3dd/131007_devstack_export.log : SUCCESS in 16m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/880a5303023d45e58b47986b601a3e91/user_001.log : SUCCESS in 21m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/aaaac47239274ba38172ebabe5ac1031/131007_devstack_export.log : SUCCESS in 14m 52s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/4/check/real-db-upgrade_nova_percona_user_001:th-percona/d75418bc663944a59e7341dd35e0befd/user_001.log : SUCCESS in 25m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/38b9d23c7ad1471784af8ae30aba3aa4/user_002.log : SUCCESS in 18m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/4/check/real-db-upgrade_nova_percona_user_002:th-percona/1fbafc82a1bd42a1969a5b384eae8f7a/user_002.log : SUCCESS in 18m 37s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"ddb9c187c973b2cf37ba8dcedefd1f1ade7cfc59","author":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"date":"2014-09-05 10:43:07.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6a10f09301570f68434b2531cdc91faa6f262cf","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-05 10:46:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3b4b62e6969873ea9a0e90b78c5698911bc2ca4a","author":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"date":"2014-09-05 10:54:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b13d6bdb198bf5f6ce6b92e7a362a68fc1a34d0c","author":{"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},"date":"2014-09-05 10:55:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b53890b55c8b318d55ee4582cb41c2c25f24707e","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-05 10:56:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6142127a91e189b75f14ab450c0a5f3756c64226","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-05 10:57:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac4406d95631e74a3f282458067530bc7c9b2954","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-05 10:58:43.000000000","message":"Patch Set 5:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"2f1f5561c0a88411dd2f65dc363137f8125d3034","author":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"date":"2014-09-05 10:59:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6f316631e5c2cd6090b63f1ccdcc16a37290c802","author":{"_account_id":10614,"name":"Alex Frolov","email":"afrolov@mirantis.com","username":"afrolov"},"date":"2014-09-05 11:02:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"65fb0c8899bf1d9da4bfd0c032f9b01484e6db1a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-05 11:12:34.000000000","message":"Patch Set 5: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/5/25037/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"8e4e6c9fa3d13bca5fb2a978a0bc3d75181018d3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-05 11:19:39.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e0e78b2097504895ad522be76f7a7b56/131007_devstack_export.log : SUCCESS in 15m 25s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e9ab307cd2af4f1cb9d5a47455cdb09b/user_001.log : SUCCESS in 22m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/63677df085ae4f56b46c271ff0a7255e/131007_devstack_export.log : SUCCESS in 14m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/5/check/real-db-upgrade_nova_percona_user_001:th-percona/6cc0cb4416cf4668a80909ed6f43fb3c/user_001.log : SUCCESS in 21m 25s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/be503c9cdafe4b6d98c2c803953585f9/user_002.log : SUCCESS in 17m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/5/check/real-db-upgrade_nova_percona_user_002:th-percona/09bb3c2208a142e491033dfdca092416/user_002.log : SUCCESS in 17m 30s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"1d8b18a8732f9e0053d016239cfce2dd68e83033","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 12:10:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/5/check/gate-nova-pep8/1328a94 : SUCCESS in 8m 33s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/5/check/gate-nova-docs/0548941/doc/build/html/ : SUCCESS in 22m 38s\n- gate-nova-python26 http://logs.openstack.org/75/118675/5/check/gate-nova-python26/6136efd : FAILURE in 1h 10m 15s\n- gate-nova-python27 http://logs.openstack.org/75/118675/5/check/gate-nova-python27/e142931 : FAILURE in 21m 36s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/5/check/check-tempest-dsvm-full/1890c75 : FAILURE in 15m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/5/check/check-tempest-dsvm-postgres-full/227cc68 : FAILURE in 13m 20s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/5/check/check-tempest-dsvm-neutron-full/df1031e : SUCCESS in 53m 07s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/5/check/check-tempest-dsvm-neutron-heat-slow/925ebde : SUCCESS in 24m 01s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/5/check/check-grenade-dsvm/7c9be95 : SUCCESS in 47m 02s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/5/check/gate-tempest-dsvm-large-ops/32c87d2 : FAILURE in 15m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/5/check/gate-tempest-dsvm-neutron-large-ops/7c11137 : SUCCESS in 20m 50s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/5/check/check-devstack-dsvm-cells/1185e9a : FAILURE in 10m 55s\n- gate-nova-pylint http://logs.openstack.org/75/118675/5/check/gate-nova-pylint/4102a5b : FAILURE in 25m 11s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/5/check/check-grenade-dsvm-partial-ncpu/71250e9 : FAILURE in 42m 37s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/75/118675/5/check/check-tempest-dsvm-virtual-ironic-nv/4084c0a : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0dc385229f361251b53cd911c7c56c9754df0348","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-08 11:48:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/118675/1","accounts_in_message":[],"_revision_number":1},{"id":"2bb2658953848818cae5192ae79aeae1485a4750","author":{"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},"date":"2014-09-08 13:59:27.000000000","message":"Patch Set 5: Code-Review-1\n\nNeed to rebase.","accounts_in_message":[],"_revision_number":5},{"id":"8816168a70a2888eee79b1df39ce2eccd63f260c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-12 12:27:52.000000000","message":"Patch Set 5:\n\nBuild failed. Logs: http://208.91.1.172/logs/118675/5 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":5},{"id":"57468449b8caae17a8a64d80efb6fe09a4bfbe37","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2014-09-13 11:11:38.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\none nit inline","accounts_in_message":[],"_revision_number":5},{"id":"0b988e4515efce1a2f9ae0fa5f147092a73082ec","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-16 11:51:26.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"deb5ff889250876718dbad9b9726ab3d15386928","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-16 12:14:31.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b02c32c84d0c4900998edad925c3e201/131007_devstack_export.log : SUCCESS in 15m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2a08c04b35e34e77917d6e02f9d8a075/user_001.log : SUCCESS in 21m 53s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/acff6d27ef834cc28201ed2601a30aec/131007_devstack_export.log : SUCCESS in 18m 32s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/6/check/real-db-upgrade_nova_percona_user_001:th-percona/c502756611b347428af26222f39910e5/user_001.log : SUCCESS in 22m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5f4f416acc404064b7a6b19d5e54b136/user_002.log : SUCCESS in 18m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/6/check/real-db-upgrade_nova_percona_user_002:th-percona/05a9cdbbddd44bc3a56d80a0498a82a4/user_002.log : SUCCESS in 18m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"84e9827cc481e9835f5cd768a64434c354e92bb3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-16 12:21:01.000000000","message":"Patch Set 6:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/6/26286/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"c4359d8dd22fc1775ac25b62554a256d8d0cd62e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-16 13:22:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/6/ : Looks good to me in 1h 30m 19s","accounts_in_message":[],"_revision_number":6},{"id":"2bc0be6a55f20fcfbb429c83ae940ec86d0e1474","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 13:58:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/6/check/gate-nova-pep8/019ede8 : SUCCESS in 9m 40s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/6/check/gate-nova-docs/a1a4535/doc/build/html/ : SUCCESS in 12m 20s\n- gate-nova-python26 http://logs.openstack.org/75/118675/6/check/gate-nova-python26/cbc32e5 : FAILURE in 14m 58s\n- gate-nova-python27 http://logs.openstack.org/75/118675/6/check/gate-nova-python27/2756359 : FAILURE in 50m 12s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/6/check/check-tempest-dsvm-full/d9bb849 : FAILURE in 14m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/6/check/check-tempest-dsvm-postgres-full/bd6b74b : FAILURE in 12m 57s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/6/check/check-tempest-dsvm-neutron-full/1c760e7 : SUCCESS in 1h 06m 31s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/6/check/check-tempest-dsvm-neutron-heat-slow/51f43ea : SUCCESS in 24m 01s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/6/check/check-grenade-dsvm/4a54812 : SUCCESS in 39m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/6/check/gate-tempest-dsvm-large-ops/8ad2669 : FAILURE in 14m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/6/check/gate-tempest-dsvm-neutron-large-ops/c18f75d : SUCCESS in 23m 13s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/6/check/check-devstack-dsvm-cells/691297a : FAILURE in 14m 42s\n- gate-nova-pylint http://logs.openstack.org/75/118675/6/check/gate-nova-pylint/9da7c28 : FAILURE in 21m 43s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/6/check/check-grenade-dsvm-partial-ncpu/4613165 : SUCCESS in 44m 49s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/75/118675/6/check/check-tempest-dsvm-virtual-ironic-nv/3acd5ce : SUCCESS in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d978d81b95e6917d1ac8d62f0f10bb70cc2c0f4f","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-16 15:44:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3343f386e45ffc8ce90c9228a2d87de828d7ef30","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-16 15:46:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c7e3297ec1a0f1553d4bf8d283eed80eade37eb","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-16 16:29:19.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a050614074cf42a0918109418031fa7d/131007_devstack_export.log : SUCCESS in 17m 47s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/590fb5354fca4753b4d215f8c36865e1/user_001.log : SUCCESS in 26m 01s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/66fd07ee251e4c9f828a33db0b7cb19b/131007_devstack_export.log : SUCCESS in 15m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/7/check/real-db-upgrade_nova_percona_user_001:th-percona/bb9d47cf917a425a9bc984a355d0ffc0/user_001.log : SUCCESS in 23m 44s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/43a311787bfd4c5083fdbc14e23e7698/user_002.log : SUCCESS in 20m 53s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/7/check/real-db-upgrade_nova_percona_user_002:th-percona/63475ba5da034b32af58a1e167f114da/user_002.log : SUCCESS in 17m 39s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"2f76c56f4a629d23b830cecc284dcfdd95c5a32c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-16 16:50:17.000000000","message":"Patch Set 7:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/7/26359/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":7},{"id":"a39b5db9d261506340c96901a13f6d81801e4464","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-16 18:24:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/7/ : Looks good to me in 1h 54m 08s","accounts_in_message":[],"_revision_number":7},{"id":"82f091471ae8abd2c9dfe852687488070fab12cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 19:52:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/7/check/gate-nova-pep8/3484407 : SUCCESS in 9m 33s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/7/check/gate-nova-docs/2580f4f/doc/build/html/ : SUCCESS in 11m 57s\n- gate-nova-python26 http://logs.openstack.org/75/118675/7/check/gate-nova-python26/5c6fc72 : FAILURE in 14m 29s\n- gate-nova-python27 http://logs.openstack.org/75/118675/7/check/gate-nova-python27/67166b3 : FAILURE in 17m 24s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/7/check/check-tempest-dsvm-full/237fcbd : FAILURE in 13m 49s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/7/check/check-tempest-dsvm-postgres-full/d0ae7ea : FAILURE in 13m 00s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/7/check/check-tempest-dsvm-neutron-full/ec86e71 : SUCCESS in 1h 06m 05s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/7/check/check-tempest-dsvm-neutron-heat-slow/8b97950 : SUCCESS in 24m 06s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/7/check/check-grenade-dsvm/2ab8eea : SUCCESS in 45m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/7/check/gate-tempest-dsvm-large-ops/a60398e : FAILURE in 14m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/7/check/gate-tempest-dsvm-neutron-large-ops/50001ae : SUCCESS in 23m 33s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/7/check/check-devstack-dsvm-cells/566d543 : FAILURE in 13m 45s\n- gate-nova-pylint http://logs.openstack.org/75/118675/7/check/gate-nova-pylint/381c0ac : FAILURE in 16m 10s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/7/check/check-grenade-dsvm-partial-ncpu/d358d8f : SUCCESS in 45m 42s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/75/118675/7/check/check-tempest-dsvm-ironic-pxe_ssh-nv/c0effdd : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"731ddda150f2d5d51460c72c511dc04cb0f73ff5","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2014-09-16 21:24:42.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e41bf1c5142c4e030a671e8d8138affe01012629","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-17 16:31:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"27cf3af2c968d40bdbd571e31d0cbaff6014b19a","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-17 16:32:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"69eaa6ed29973ef91cb02e5e4acbf8caea01810a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-17 16:51:16.000000000","message":"Patch Set 8: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/8/26539/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":8},{"id":"8654112d8adb8fdcf5acd32db97fd98ba169d1cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-17 18:43:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/8/check/gate-nova-pep8/13d7920 : SUCCESS in 11m 58s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/8/check/gate-nova-docs/77dd87c/doc/build/html/ : SUCCESS in 17m 27s\n- gate-nova-python26 http://logs.openstack.org/75/118675/8/check/gate-nova-python26/1c43886 : SUCCESS in 16m 20s\n- gate-nova-python27 http://logs.openstack.org/75/118675/8/check/gate-nova-python27/9a35aed : SUCCESS in 15m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-full/ab1d840 : FAILURE in 11m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-postgres-full/2f5f2fb : FAILURE in 14m 20s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-neutron-full/cf1f92e : FAILURE in 51m 31s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-neutron-heat-slow/19c5978 : SUCCESS in 19m 08s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/8/check/check-grenade-dsvm/551ec4a : SUCCESS in 41m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/8/check/gate-tempest-dsvm-large-ops/4438859 : FAILURE in 14m 04s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/8/check/gate-tempest-dsvm-neutron-large-ops/23b0d55 : SUCCESS in 23m 18s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/8/check/check-devstack-dsvm-cells/4171bfb : FAILURE in 15m 38s\n- gate-nova-pylint http://logs.openstack.org/75/118675/8/check/gate-nova-pylint/0b298e4 : FAILURE in 24m 19s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/8/check/check-grenade-dsvm-partial-ncpu/e85dc07 : SUCCESS in 45m 04s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-ironic-pxe_ssh-nv/9e8bf89 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2d2505000a8f2e22e2327a88fd8798c6c63d6147","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-17 18:43:28.000000000","message":"Patch Set 8:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/118675/8/ : Test run failed. in 2h 11m 05s","accounts_in_message":[],"_revision_number":8},{"id":"18f04f0de21e828fe77e8ee7ef728e24e21d08e5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-18 13:03:05.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9f32d14530404b878d84de91d7498c82/131007_devstack_export.log : SUCCESS in 14m 48s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9fdfed3e26d943668715779ec96f07d0/user_001.log : SUCCESS in 26m 17s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fdb47450127f44c0bd785fd417f30e6e/131007_devstack_export.log : SUCCESS in 16m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/8/check/real-db-upgrade_nova_percona_user_001:th-percona/9a3e5d2133d5446aad51bde83920b0b8/user_001.log : SUCCESS in 30m 09s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6614fd58fa7747429e0fd2df302e54c4/user_002.log : SUCCESS in 22m 30s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/8/check/real-db-upgrade_nova_percona_user_002:th-percona/9febc57a56e74cd28dc5d6e4bb6296f9/user_002.log : SUCCESS in 17m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"c4ae56e7f7523892dd95afe79f14af15626118a1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-22 21:32:48.000000000","message":"Patch Set 8:\n\nBuild failed. Logs: http://208.91.1.172/logs/118675/8 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":8},{"id":"45c5cd5b98c6a530f62e1a676d0b33458463d69e","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-09-24 20:05:53.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3b9657aab5153354348b40f67628de934c107e6b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-24 22:18:08.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/8/check/gate-nova-pep8/56a6b21 : SUCCESS in 11m 50s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/8/check/gate-nova-docs/a0a9349/doc/build/html/ : SUCCESS in 23m 49s\n- gate-nova-python26 http://logs.openstack.org/75/118675/8/check/gate-nova-python26/21bd457 : SUCCESS in 16m 13s\n- gate-nova-python27 http://logs.openstack.org/75/118675/8/check/gate-nova-python27/b3770c8 : SUCCESS in 19m 43s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-full/28f9eb8 : FAILURE in 16m 22s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-postgres-full/4bf7b4c : FAILURE in 15m 39s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-neutron-full/fb3a61b : SUCCESS in 1h 07m 50s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-neutron-heat-slow/90f076d : SUCCESS in 25m 37s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/8/check/check-grenade-dsvm/238cb33 : SUCCESS in 47m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/8/check/gate-tempest-dsvm-large-ops/b12c339 : FAILURE in 14m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/8/check/gate-tempest-dsvm-neutron-large-ops/bc9cbe5 : SUCCESS in 24m 02s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/8/check/check-devstack-dsvm-cells/583231a : FAILURE in 16m 04s\n- gate-nova-pylint http://logs.openstack.org/75/118675/8/check/gate-nova-pylint/a10f294 : FAILURE in 22m 15s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/8/check/check-grenade-dsvm-partial-ncpu/1d5cf9a : SUCCESS in 39m 50s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/75/118675/8/check/check-tempest-dsvm-ironic-pxe_ssh-nv/ac2b088 : SUCCESS in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aac1cb8a3a802c5709ca08cd361a33a85a13ea16","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-24 23:40:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/8/ : Looks good to me in 1h 16m 39s","accounts_in_message":[],"_revision_number":8},{"id":"f8562b5256d639c73681e87a6eb99486383fc901","author":{"_account_id":8412,"name":"Mike Durnosvistov","email":"glacierr.dev@gmail.com","username":"mdurnosvistov"},"date":"2014-10-01 11:36:00.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"946d1036657400400bb05d9978e3e5f5dfaeef18","author":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"date":"2014-10-06 10:14:38.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0230dbee66c26dd4f162dd8f8126e82abf056d8f","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-10-08 07:16:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9914143cc2071579edf8b2bcfcdbbf0e38c57fe0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-10-08 07:32:35.000000000","message":"Patch Set 9: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/9/27830/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":9},{"id":"1141822c1c18ce02b102e4d9ede246d9e431bf84","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-10-08 07:41:04.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/df7d5427aff94ed0b9950a68b2908753/131007_devstack_export.log : SUCCESS in 16m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/db389fe0bfbf42ca9fc426127214a640/user_001.log : SUCCESS in 22m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/968d2808c4fd4b0bbe7fe4c39c7eb9c0/131007_devstack_export.log : SUCCESS in 19m 48s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/9/check/real-db-upgrade_nova_percona_user_001:th-percona/17f2c60d03db49c482baaeda2f8c04fe/user_001.log : SUCCESS in 23m 43s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1fd51781752e419f9524142b58590531/user_002.log : SUCCESS in 19m 31s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/9/check/real-db-upgrade_nova_percona_user_002:th-percona/16dd04b2b66d4c8182889b9fd1c617ac/user_002.log : SUCCESS in 20m 24s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"9907ce5bc2894a06b069e0eb06d9ef67ba2cf917","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-08 08:20:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/9/check/gate-nova-pep8/07eccf1 : SUCCESS in 9m 07s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/9/check/gate-nova-docs/698ef3b/doc/build/html/ : SUCCESS in 7m 06s\n- gate-nova-python26 http://logs.openstack.org/75/118675/9/check/gate-nova-python26/9978f45 : SUCCESS in 16m 47s\n- gate-nova-python27 http://logs.openstack.org/75/118675/9/check/gate-nova-python27/c2bdcac : SUCCESS in 19m 34s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/9/check/check-tempest-dsvm-full/ac4b056 : FAILURE in 14m 19s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/9/check/check-tempest-dsvm-postgres-full/250c978 : FAILURE in 13m 30s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/9/check/check-tempest-dsvm-neutron-full/45fe554 : SUCCESS in 1h 02m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/75/118675/9/check/check-tempest-dsvm-neutron-heat-slow/ce8bdc3 : SUCCESS in 22m 26s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/9/check/check-grenade-dsvm/b93c4ec : SUCCESS in 47m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/9/check/gate-tempest-dsvm-large-ops/872a43b : FAILURE in 20m 15s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/9/check/gate-tempest-dsvm-neutron-large-ops/a92070e : SUCCESS in 17m 42s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/9/check/check-devstack-dsvm-cells/e7ff7c2 : FAILURE in 11m 37s\n- gate-nova-pylint http://logs.openstack.org/75/118675/9/check/gate-nova-pylint/4c6430b : SUCCESS in 21m 16s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/9/check/check-grenade-dsvm-partial-ncpu/d61af4a : SUCCESS in 43m 05s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/75/118675/9/check/check-tempest-dsvm-ironic-pxe_ssh-nv/e7920dc : SUCCESS in 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f74232ded14b1d95627fb586c7937ce55a7ab6b1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-10-09 06:43:07.000000000","message":"Patch Set 9:\n\nBuild failed. Logs: http://208.91.1.172/logs/118675/9 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":9},{"id":"a17d5853bf5b5100f26bc0d24e7d1db955b65b87","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-11-20 15:19:35.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":9},{"id":"bce49d5ce03892a738da3baf755e899da68b50e9","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-11-20 15:19:35.000000000","message":"Patch Set 9:\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":9},{"id":"45be1ae443ac68aaa26552c1a28b1fd6c079600b","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 09:11:01.000000000","message":"Restored","accounts_in_message":[],"_revision_number":9},{"id":"f26e9f4b07e61757c59fc770dde6a5d3f462ba07","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-12-01 09:11:17.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"36d2ed6455c538ad4d6592760d588f74080fbde7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-01 09:11:22.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"4e0e07d6ad650e4fd0c8b07b5f158f6fcbac545f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-01 09:11:37.000000000","message":"Patch Set 9:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"df737b99d0664e0aec1ff976ca8ab7ab2b38e439","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-01 09:11:40.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"d5d4f3d3f2ded574056e9fbdc58b479ff6d7ef45","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 09:57:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"21c188469016ee149537cc00467748d6348418b9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-01 10:22:30.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bbce163def4f4b9e9f9bda35e4729475/131007_devstack_export.log : SUCCESS in 19m 38s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6288beffb37b46448bb197f7ade65d85/user_001.log : SUCCESS in 22m 01s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d357c1d02aec46c48310690042310b89/131007_devstack_export.log : SUCCESS in 17m 31s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/10/check/real-db-upgrade_nova_percona_user_001:th-percona/b3a1c6f755a942809d1ba6a8c74e8f74/user_001.log : SUCCESS in 23m 26s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/15221a9b1b294d53a124cd83ee537d50/user_002.log : SUCCESS in 17m 41s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/10/check/real-db-upgrade_nova_percona_user_002:th-percona/a8b11f01a60e40889e0cf533d0735d83/user_002.log : SUCCESS in 18m 17s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"ba2afd3f6f127c561cc2080d8a9bdf6b439c84e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-01 10:37:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/10/check/gate-nova-pep8/8cd15a7 : FAILURE in 1m 02s\n- gate-nova-docs http://logs.openstack.org/75/118675/10/check/gate-nova-docs/96b2cbc : FAILURE in 1m 03s\n- gate-nova-python27 http://logs.openstack.org/75/118675/10/check/gate-nova-python27/7b43450 : FAILURE in 1m 04s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-full/1862670 : SUCCESS in 29m 41s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-postgres-full/6b8cc3c : SUCCESS in 28m 38s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-neutron-full/6c317eb : SUCCESS in 39m 06s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/10/check/check-grenade-dsvm/4bcb30e : SUCCESS in 26m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/10/check/gate-tempest-dsvm-large-ops/3bae46a : SUCCESS in 17m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/10/check/gate-tempest-dsvm-neutron-large-ops/b1cb267 : SUCCESS in 16m 20s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/10/check/check-devstack-dsvm-cells/c7a4954 : SUCCESS in 14m 16s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/10/check/gate-nova-tox-functional/d37d148 : FAILURE in 1m 25s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/10/check/check-grenade-dsvm-partial-ncpu/de2154d : SUCCESS in 25m 51s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-ironic-pxe_ssh/0e5655a : SUCCESS in 26m 25s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-docker/d367f42 : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1d14fa658fd11dd4583db9911b0f30210b555b45","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-12-01 10:40:30.000000000","message":"Patch Set 10: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/10/825/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":10},{"id":"9fac1a75d5b4aa784ebebc2c28d53fd8e3bc9de1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-12-01 10:47:32.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/118675/10/check-ibm-tempest-dsvm-full/05e68c6 : SUCCESS in 43m 23s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/118675/10/check-ibm-tempest-dsvm-postgres-full/10468a6 : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"34bc2386f1d85185fc4fd50025837766b08550ac","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-01 11:07:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/10/ : Looks good to me in 1h 09m 35s","accounts_in_message":[],"_revision_number":10},{"id":"785e2a9a49f89033c3f11ab74bea1f6151665a6d","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2014-12-01 12:57:37.000000000","message":"Patch Set 10: Code-Review+1\n\nFailure seems to be due to a problem with gates.\n\nlgtm","accounts_in_message":[],"_revision_number":10},{"id":"8eb610feaef5b4d468cf5834524f09bf802fae04","author":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"date":"2014-12-01 13:09:41.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c3b72cb47f7b5a61c6a506267033f1a0fce6c561","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 13:21:53.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"9948546179acceaee6e750a464753f1cd6a15f36","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-01 14:02:39.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/10/check/gate-nova-pep8/8c5524e : SUCCESS in 11m 21s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/10/check/gate-nova-docs/425c2e2/doc/build/html/ : SUCCESS in 9m 14s\n- gate-nova-python27 http://logs.openstack.org/75/118675/10/check/gate-nova-python27/04ade8d : SUCCESS in 13m 24s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-full/8c0edb8 : SUCCESS in 36m 52s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-postgres-full/9b4bdd6 : SUCCESS in 33m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-neutron-full/595fe03 : SUCCESS in 33m 17s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/10/check/check-grenade-dsvm/41bb754 : SUCCESS in 24m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/10/check/gate-tempest-dsvm-large-ops/8cf704a : SUCCESS in 18m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/10/check/gate-tempest-dsvm-neutron-large-ops/f41150f : SUCCESS in 20m 40s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/10/check/check-devstack-dsvm-cells/60759b1 : SUCCESS in 12m 26s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/10/check/gate-nova-tox-functional/16e1319 : FAILURE in 11m 41s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/10/check/check-grenade-dsvm-partial-ncpu/950d5af : SUCCESS in 28m 58s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-ironic-pxe_ssh/4991167 : SUCCESS in 24m 03s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/10/check/check-tempest-dsvm-docker/9e10240 : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6a17d9f95e3d2fd4e1c143de55f3b320ba3a5e7d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-01 14:41:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/10/ : Looks good to me in 1h 19m 25s","accounts_in_message":[],"_revision_number":10},{"id":"236f38d52908b7560d0950f8dd451f3b2502cb89","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 14:45:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ff0235dc554878e0008e203f9e9e2d34443d99ff","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-01 15:12:21.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/69341271f8de4b92bb17a6409b16d128/131007_devstack_export.log : SUCCESS in 19m 10s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4643ceb17ba04a06afe2e7c4acc54dbb/user_001.log : SUCCESS in 25m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9f7fa13a3f1946b194fec7fe48de2aae/131007_devstack_export.log : SUCCESS in 18m 46s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/11/check/real-db-upgrade_nova_percona_user_001:th-percona/da85986682b84408a906ecc5d5215e7e/user_001.log : SUCCESS in 25m 12s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/850b7248a4de4548ad90b27a8ca69cfe/user_002.log : SUCCESS in 18m 52s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/11/check/real-db-upgrade_nova_percona_user_002:th-percona/c2885126bd294d549e68e47917799316/user_002.log : SUCCESS in 19m 22s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"50060a57c7f86a602b168e4be10f22799da4bde9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-01 15:27:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/11/check/gate-nova-pep8/2d5e28e : SUCCESS in 10m 29s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/11/check/gate-nova-docs/e1c4ecf/doc/build/html/ : SUCCESS in 7m 34s\n- gate-nova-python27 http://logs.openstack.org/75/118675/11/check/gate-nova-python27/1de6bf5 : SUCCESS in 9m 35s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/11/check/check-tempest-dsvm-full/db9b56d : SUCCESS in 29m 22s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/11/check/check-tempest-dsvm-postgres-full/1f3a93b : SUCCESS in 39m 17s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/11/check/check-tempest-dsvm-neutron-full/e3567e5 : SUCCESS in 41m 40s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/11/check/check-grenade-dsvm/a070ef8 : SUCCESS in 27m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/11/check/gate-tempest-dsvm-large-ops/f796958 : SUCCESS in 23m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/11/check/gate-tempest-dsvm-neutron-large-ops/1f7cbc2 : SUCCESS in 19m 23s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/11/check/check-devstack-dsvm-cells/d7ce756 : SUCCESS in 13m 37s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/11/check/gate-nova-tox-functional/78b4846 : SUCCESS in 11m 01s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/11/check/check-grenade-dsvm-partial-ncpu/6af883d : SUCCESS in 28m 14s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/11/check/check-tempest-dsvm-ironic-pxe_ssh/52aef89 : SUCCESS in 27m 06s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/11/check/check-tempest-dsvm-docker/dbaca77 : SUCCESS in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"96905a114c1612e7b4dfd1ca66770a6bb9fadea0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-12-01 15:29:07.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/118675/11/check-ibm-tempest-dsvm-full/3a6f69e : SUCCESS in 42m 48s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/118675/11/check-ibm-tempest-dsvm-postgres-full/6dcc35c : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5caa99f2e85cabe1a1f3c121c71594638fddb11b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-12-01 15:33:08.000000000","message":"Patch Set 11: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/11/855/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":11},{"id":"08cb4410c6a0c08a7f4b6744e78b8798c7d9c7ea","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-01 15:41:07.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/11/ : Looks good to me in 54m 39s","accounts_in_message":[],"_revision_number":11},{"id":"dabde370a9d79ea5c8d4b0a216c05a557caffa08","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 15:49:22.000000000","message":"Patch Set 11:\n\n/16 network creates in 24s instead of 1m20s, /13 network creates in 3m25s instead of 14m28s","accounts_in_message":[],"_revision_number":11},{"id":"cd5bf7d1ffb87ee9b6427f61171b234f70d24b6f","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2014-12-01 15:53:08.000000000","message":"Patch Set 11: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":11},{"id":"617f03bf35e76d78e8ba6ff48253084a23faf84d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-12-01 16:31:45.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nHi Pavel!\n\nCouple minor things to address inline, but overall, excellent work. Thank you!\n\n-jay","accounts_in_message":[],"_revision_number":11},{"id":"3ade3f11dbc1dfe14e58d14f33bb06655f3fd5c6","author":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"date":"2014-12-01 16:59:59.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)\n\nI don\u0027t have anything beyond what Jay suggested. Just a quick question inline, but otherwise looks pretty good to me :-)","accounts_in_message":[],"_revision_number":11},{"id":"e6414d539b9f0928143dbfad4c1d7d4527b28fec","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 17:10:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"dc86d194557105cf1250c32e90060161a75e51e8","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-01 17:21:51.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d2c0cdb70550e3501a58f23596b60b1e8b4239db","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-01 17:32:16.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/c0c12a1a35bd401b8469114ab3b43a58/131007_devstack_export.log : SUCCESS in 16m 59s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/20f4eb8d9a2a4d9885662315eaf788e1/user_001.log : SUCCESS in 20m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/68d1ca3fb19842e8923e970fca9b8a98/131007_devstack_export.log : SUCCESS in 19m 39s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_user_001:th-percona/e27d469b6fa241588ed8fbb2a69d2393/user_001.log : SUCCESS in 18m 12s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/fbfe1e9c65fd4c0c833372be72a09dc8/user_002.log : SUCCESS in 17m 28s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_user_002:th-percona/21912305e33344ed812fff7875513451/user_002.log : SUCCESS in 17m 22s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"e3519f2ef795f4183f9027e61a0ddbaad36145a4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-12-01 17:50:15.000000000","message":"Patch Set 12: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/75/118675/12/877/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":12},{"id":"31743d9c94bcb92a762c30d739ab75e6b7c30115","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-01 17:50:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/12/check/gate-nova-pep8/dc8660e : SUCCESS in 11m 20s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/12/check/gate-nova-docs/3095cca/doc/build/html/ : SUCCESS in 7m 08s\n- gate-nova-python27 http://logs.openstack.org/75/118675/12/check/gate-nova-python27/3b3fc1b : SUCCESS in 9m 14s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-full/357009a : SUCCESS in 30m 39s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-postgres-full/ca3a8d2 : SUCCESS in 38m 29s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-neutron-full/09180ab : SUCCESS in 39m 52s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm/9084c51 : SUCCESS in 31m 23s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-large-ops/0dea54b : SUCCESS in 21m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-neutron-large-ops/37c3b3d : SUCCESS in 20m 16s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/12/check/check-devstack-dsvm-cells/c1ea353 : SUCCESS in 12m 05s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/12/check/gate-nova-tox-functional/2fa19cd : SUCCESS in 14m 54s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm-partial-ncpu/b97a58e : SUCCESS in 23m 33s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-ironic-pxe_ssh/f06ac4c : SUCCESS in 23m 33s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-docker/001ef5c : SUCCESS in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6cbf63fde080836e69bb358ddd7e81a9e4144c88","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-12-01 17:52:19.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/118675/12/check-ibm-tempest-dsvm-full/4a640b4 : SUCCESS in 40m 18s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/118675/12/check-ibm-tempest-dsvm-postgres-full/2a6dc8d : SUCCESS in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c60918b5984c85f3bb4ed8c4e5db77ef45675d05","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-01 18:31:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/12/ : Looks good to me in 1h 20m 59s","accounts_in_message":[],"_revision_number":12},{"id":"b8c4f7e876a367e09c1fd96ef9b52326355e133f","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-12-02 00:22:47.000000000","message":"Patch Set 12: Code-Review+2\n\nlgtm, thx for fixing that up, Pavel.","accounts_in_message":[],"_revision_number":12},{"id":"22d9740d73f0225e2f6e6189b5c051752e5dcba4","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2014-12-02 12:26:09.000000000","message":"Patch Set 12: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":12},{"id":"0cd4301948ba1aefad3f8b09d6dc30b982b6b8b7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-12-02 14:58:04.000000000","message":"Patch Set 12:\n\nBuild failed. Logs: http://208.91.1.172/logs/118675/12 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":12},{"id":"69503edd99b8be8855bca570120f97dd99c81c9f","author":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"date":"2014-12-02 15:03:10.000000000","message":"Patch Set 12: Code-Review+1\n\ngood work!","accounts_in_message":[],"_revision_number":12},{"id":"2780503f27900e795db08dde3f79208cb5845d29","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-12-02 15:08:56.000000000","message":"Patch Set 12: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0e62734dbbb1578f4052c0d8f2fb3011a21d2f63","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-02 15:39:13.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8a927ac6f905e49de2f5a508e8de31f1eb6cf309","author":{"_account_id":11531,"name":"warewang","email":"libra1588@126.com","username":"warewang"},"date":"2014-12-03 02:15:38.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"ddb4ba6512e116f5780a394f77ba628c8b149980","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-12-03 11:04:35.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(2 comments)\n\nLooks good, while I don\u0027t doubt removing the python loop and using SQL is faster, I am curious how much faster.","accounts_in_message":[],"_revision_number":12},{"id":"7465f62331d7fae763498826b8fef9d7388c6065","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 11:04:38.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"51eb954d15fdb4015aa59614df437fd3e1bd54b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 11:49:09.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/12/gate/gate-nova-docs/de319c4/doc/build/html/ : SUCCESS in 6m 36s\n- gate-nova-pep8 http://logs.openstack.org/75/118675/12/gate/gate-nova-pep8/0496509 : SUCCESS in 8m 24s\n- gate-nova-python27 http://logs.openstack.org/75/118675/12/gate/gate-nova-python27/397cb3b : FAILURE in 10m 27s\n- gate-tempest-dsvm-full http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-full/96c81c9 : SUCCESS in 36m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-postgres-full/494b5ff : SUCCESS in 32m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-neutron-full/75628d7 : SUCCESS in 35m 30s\n- gate-grenade-dsvm http://logs.openstack.org/75/118675/12/gate/gate-grenade-dsvm/c1ee1dc : SUCCESS in 28m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-large-ops/f50838c : SUCCESS in 23m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-neutron-large-ops/9a573c4 : SUCCESS in 21m 28s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/12/gate/gate-nova-tox-functional/44e368b : SUCCESS in 11m 13s\n- gate-devstack-dsvm-cells http://logs.openstack.org/75/118675/12/gate/gate-devstack-dsvm-cells/b1c2fee : SUCCESS in 15m 45s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/12/gate/gate-grenade-dsvm-partial-ncpu/4c48104 : SUCCESS in 27m 23s","accounts_in_message":[],"_revision_number":12},{"id":"1f0e97c41bc691f5f020d8b343a4ca39804e3fc0","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-12-03 11:49:25.000000000","message":"Patch Set 12:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27: https://bugs.launchpad.net/bugs/1398624\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":12},{"id":"94d73c8bfb0765d9e3dcc0beb1bdd71d7d66560f","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-03 11:58:20.000000000","message":"Patch Set 12:\n\nrecheck bug 1398624","accounts_in_message":[],"_revision_number":12},{"id":"82ec322a298de2a19ffea66fcb7de05db49e37f8","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-03 12:22:12.000000000","message":"Patch Set 12:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/80487e580f9e4aaf8c7113c84b39ab75/index.html : SUCCESS in 21m 57s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/523a72c3bf6d4d76b2dc24a186c72e01/index.html : SUCCESS in 23m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/21154e012e084ea2a6caf160f15f0279/index.html : SUCCESS in 18m 56s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_user_001:th-percona/0cbe2ac4a9ad478195449be2c566e664/index.html : SUCCESS in 21m 07s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/b96cd336f46d45e4b8031932b7de72dc/index.html : SUCCESS in 18m 19s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_user_002:th-percona/867cadb04ad044fc9a4d37ed16e1d78c/index.html : SUCCESS in 18m 16s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"2a3d2a229de7ebf9244c5fdb5b48ca9df632e667","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 12:35:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/12/check/gate-nova-pep8/cd9f0bd : SUCCESS in 8m 30s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/12/check/gate-nova-docs/126725f/doc/build/html/ : SUCCESS in 8m 41s\n- gate-nova-python27 http://logs.openstack.org/75/118675/12/check/gate-nova-python27/a59889f : FAILURE in 10m 04s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-full/f2d788a : SUCCESS in 33m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-postgres-full/67669de : SUCCESS in 35m 21s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-neutron-full/d449e7e : SUCCESS in 37m 16s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm/67fbe25 : SUCCESS in 29m 09s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-large-ops/8b81589 : SUCCESS in 23m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-neutron-large-ops/e0c14ee : SUCCESS in 17m 34s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/12/check/check-devstack-dsvm-cells/327f666 : SUCCESS in 14m 56s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/12/check/gate-nova-tox-functional/27bd93b : SUCCESS in 10m 59s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm-partial-ncpu/873806e : SUCCESS in 27m 27s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-ironic-pxe_ssh/ae02272 : SUCCESS in 26m 42s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-docker/75a345a : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7c113c8a97a3be994281ffc6cbe68693d592a174","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-12-03 12:36:06.000000000","message":"Patch Set 12:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27: https://bugs.launchpad.net/bugs/1398624\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":12},{"id":"b9b8a8e4eb0dd357b9bfac7e17c9e64feec0b192","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-03 13:08:03.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/12/ : Looks good to me in 1h 09m 24s","accounts_in_message":[],"_revision_number":12},{"id":"4fb5ff686eedc87ce75a8917befcded566214103","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-03 17:04:18.000000000","message":"Patch Set 12:\n\nrecheck bug 1398624","accounts_in_message":[],"_revision_number":12},{"id":"d4d839e7ddb1b7afec73f3d6f1fcc00450ac4cac","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-12-03 17:53:10.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c76538df3e08206f94fbed01e0c29e45eefdd5ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 18:02:31.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/12/check/gate-nova-pep8/ed69b36 : SUCCESS in 8m 22s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/12/check/gate-nova-docs/ae3eb54/doc/build/html/ : SUCCESS in 8m 39s\n- gate-nova-python27 http://logs.openstack.org/75/118675/12/check/gate-nova-python27/a520249 : FAILURE in 13m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-full/9dffbff : SUCCESS in 31m 53s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-postgres-full/f1b208a : SUCCESS in 30m 37s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-neutron-full/ed28375 : SUCCESS in 40m 02s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm/2a0ed43 : SUCCESS in 29m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-large-ops/aa22929 : SUCCESS in 24m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-neutron-large-ops/5b36cbe : SUCCESS in 23m 39s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/12/check/check-devstack-dsvm-cells/15014f3 : SUCCESS in 15m 24s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/12/check/gate-nova-tox-functional/f2dc54a : SUCCESS in 15m 26s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm-partial-ncpu/40bb80d : SUCCESS in 27m 45s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-ironic-pxe_ssh/19136bb : SUCCESS in 27m 50s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-docker/c5042ef : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bfb8c1d123a95c082a5690162741c815ef0a2121","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-03 18:02:58.000000000","message":"Patch Set 12:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7b0d05f0d0fb44a5bd60fa84112eb832/index.html : SUCCESS in 17m 42s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b820921b54ec42d5bbbb38187fac6802/index.html : SUCCESS in 19m 34s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8468edcd8997497099e9bb2bdd704045/index.html : SUCCESS in 17m 57s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_user_001:th-percona/a9dda7e5157a4205ae8e63509fad6dd2/index.html : SUCCESS in 19m 45s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0e331105ce374fcaa222e02a6f432b65/index.html : SUCCESS in 21m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/75/118675/12/check/real-db-upgrade_nova_percona_user_002:th-percona/bb30a56ce5f34eb6b639e0c7a7d43035/index.html : SUCCESS in 17m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"c0139ead3ae291ceaa241f6808cb3c2106217356","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2014-12-03 19:04:33.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c5830a4b78cc4907e05bb5704b5ea305e255d82e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 19:44:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/75/118675/12/check/gate-nova-pep8/159aa46 : SUCCESS in 11m 37s\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/12/check/gate-nova-docs/2334d3d/doc/build/html/ : SUCCESS in 9m 37s\n- gate-nova-python27 http://logs.openstack.org/75/118675/12/check/gate-nova-python27/2e1e95f : SUCCESS in 12m 42s\n- check-tempest-dsvm-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-full/90eff95 : SUCCESS in 28m 45s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-postgres-full/82ab3b5 : SUCCESS in 32m 47s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-neutron-full/e8b86b7 : SUCCESS in 34m 56s\n- check-grenade-dsvm http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm/ec3b746 : SUCCESS in 27m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-large-ops/8363ef0 : SUCCESS in 25m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/12/check/gate-tempest-dsvm-neutron-large-ops/f0cdae2 : SUCCESS in 17m 22s\n- check-devstack-dsvm-cells http://logs.openstack.org/75/118675/12/check/check-devstack-dsvm-cells/61de625 : SUCCESS in 14m 39s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/12/check/gate-nova-tox-functional/c108bce : SUCCESS in 14m 58s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/12/check/check-grenade-dsvm-partial-ncpu/0a85fe6 : SUCCESS in 28m 05s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-ironic-pxe_ssh/4a03b66 : SUCCESS in 28m 37s\n- check-tempest-dsvm-docker http://logs.openstack.org/75/118675/12/check/check-tempest-dsvm-docker/3fce555 : SUCCESS in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f25683f8dbfcf82277491526571e6ecfa609009d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 19:44:22.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"8627096656f7f08c9ff8f9eca9f4567692839c7f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-03 20:54:00.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- http://64.119.130.115/118675/12/ : Looks good to me in 1h 47m 16s","accounts_in_message":[],"_revision_number":12},{"id":"96198523575e5aa2f067752c3b44fda79f54d547","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 20:56:54.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/75/118675/12/gate/gate-nova-docs/31cbaa0/doc/build/html/ : SUCCESS in 9m 03s\n- gate-nova-pep8 http://logs.openstack.org/75/118675/12/gate/gate-nova-pep8/789094a : SUCCESS in 11m 50s\n- gate-nova-python27 http://logs.openstack.org/75/118675/12/gate/gate-nova-python27/b3902c3 : SUCCESS in 13m 42s\n- gate-tempest-dsvm-full http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-full/e73f942 : SUCCESS in 26m 25s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-postgres-full/d65d33e : SUCCESS in 39m 33s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-neutron-full/0869d09 : SUCCESS in 46m 04s\n- gate-grenade-dsvm http://logs.openstack.org/75/118675/12/gate/gate-grenade-dsvm/008fb30 : SUCCESS in 28m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-large-ops/c8e223e : SUCCESS in 23m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/75/118675/12/gate/gate-tempest-dsvm-neutron-large-ops/b2d479d : SUCCESS in 24m 09s\n- gate-nova-tox-functional http://logs.openstack.org/75/118675/12/gate/gate-nova-tox-functional/4442b41 : SUCCESS in 15m 35s\n- gate-devstack-dsvm-cells http://logs.openstack.org/75/118675/12/gate/gate-devstack-dsvm-cells/c394be8 : SUCCESS in 21m 09s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/75/118675/12/gate/gate-grenade-dsvm-partial-ncpu/8a6efe4 : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":12},{"id":"d9621fd8ac7c6afc3c94a9911e44340b8308a590","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-03 20:56:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"44f5c16f10d6b646c2e197b3b9fa7eb9ec184935","revisions":{"8668bc5ff96eafc65f21da1a9dd3c668316fa4a6":{"kind":"REWORK","_number":1,"created":"2014-09-03 16:03:58.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/1"}}},"commit":{"parents":[{"commit":"0c3d2d47cacf30bd4d4fa6d19135dd305c71f814","subject":"Move to oslo.db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c3d2d47cacf30bd4d4fa6d19135dd305c71f814"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-03 15:58:49.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-03 15:58:49.000000000","tz":240},"subject":"Optimimize \u0027fixed_ip_bulk_create\u0027 function","message":"Optimimize \u0027fixed_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) Explicit flush() was removed\n\n3) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8668bc5ff96eafc65f21da1a9dd3c668316fa4a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8668bc5ff96eafc65f21da1a9dd3c668316fa4a6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0c3d2d47cacf30bd4d4fa6d19135dd305c71f814","is_merged_in_target_branch":false,"change_id":"I0649539e071b2318ec85ed5d70259c949408e64b","change_number":101901,"patch_set_number":40,"change_status":"MERGED"}],"branch":"refs/heads/master"},"48802a869650db42a8856ac74feb542c50c890f4":{"kind":"REWORK","_number":2,"created":"2014-09-04 15:54:42.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/2"}}},"commit":{"parents":[{"commit":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","subject":"Move to oslo.db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cec8d482a68158ac608efeeef77fbfc4b43dbda1"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"subject":"Optimimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48802a869650db42a8856ac74feb542c50c890f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48802a869650db42a8856ac74feb542c50c890f4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","is_merged_in_target_branch":false,"change_id":"I0649539e071b2318ec85ed5d70259c949408e64b","change_number":101901,"patch_set_number":44,"change_status":"MERGED"}],"branch":"refs/heads/master"},"409738073179ca38c23d33deecba46c86a831ceb":{"kind":"REWORK","_number":3,"created":"2014-09-05 09:14:21.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/3"}}},"commit":{"parents":[{"commit":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","subject":"Move to oslo.db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cec8d482a68158ac608efeeef77fbfc4b43dbda1"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-05 09:13:59.000000000","tz":240},"subject":"Optimimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/409738073179ca38c23d33deecba46c86a831ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/409738073179ca38c23d33deecba46c86a831ceb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","is_merged_in_target_branch":false,"change_id":"I0649539e071b2318ec85ed5d70259c949408e64b","change_number":101901,"patch_set_number":44,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0eea5f15c754676b685a4108904a0033efeb3c87":{"kind":"REWORK","_number":4,"created":"2014-09-05 10:02:21.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/4"}}},"commit":{"parents":[{"commit":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","subject":"Move to oslo.db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cec8d482a68158ac608efeeef77fbfc4b43dbda1"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-05 10:01:53.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0eea5f15c754676b685a4108904a0033efeb3c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0eea5f15c754676b685a4108904a0033efeb3c87"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","is_merged_in_target_branch":false,"change_id":"I0649539e071b2318ec85ed5d70259c949408e64b","change_number":101901,"patch_set_number":44,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c654e450531a65474c1e03f539c5c06f8fe385c7":{"kind":"REWORK","_number":5,"created":"2014-09-05 10:56:36.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/5"}}},"commit":{"parents":[{"commit":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","subject":"Move to oslo.db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cec8d482a68158ac608efeeef77fbfc4b43dbda1"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-05 10:56:13.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c654e450531a65474c1e03f539c5c06f8fe385c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c654e450531a65474c1e03f539c5c06f8fe385c7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cec8d482a68158ac608efeeef77fbfc4b43dbda1","is_merged_in_target_branch":false,"change_id":"I0649539e071b2318ec85ed5d70259c949408e64b","change_number":101901,"patch_set_number":44,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0a4e2e0b67dc9622906ccb8654569b0140a690e4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2014-09-16 11:51:26.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/6"}}},"commit":{"parents":[{"commit":"6c36e9c411a9007def1c0748a64e2bb945ec3940","subject":"Merge \"Fix race condition in update_dhcp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c36e9c411a9007def1c0748a64e2bb945ec3940"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"Pavel Kholkin","email":"pkholkin@mirantis.com","date":"2014-09-16 11:51:25.000000000","tz":0},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a4e2e0b67dc9622906ccb8654569b0140a690e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a4e2e0b67dc9622906ccb8654569b0140a690e4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c36e9c411a9007def1c0748a64e2bb945ec3940","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"71bf50d98f1533155622804f0654f1b2df3019b0":{"kind":"REWORK","_number":7,"created":"2014-09-16 15:44:42.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/7"}}},"commit":{"parents":[{"commit":"5ab7c0a41438b9e8915f55eb0710d95bb4b6ed5c","subject":"Merge \"fix up assertEqual(None...) check to catch more cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ab7c0a41438b9e8915f55eb0710d95bb4b6ed5c"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-16 15:44:43.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/71bf50d98f1533155622804f0654f1b2df3019b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/71bf50d98f1533155622804f0654f1b2df3019b0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ab7c0a41438b9e8915f55eb0710d95bb4b6ed5c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"53310a45cb36e54e61e0e48e821e6cc3df93fe2f":{"kind":"REWORK","_number":8,"created":"2014-09-17 16:31:58.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/8"}}},"commit":{"parents":[{"commit":"c52f0a8b2432de858f5592c298d74afc9984d1b4","subject":"Merge \"Split up libvirt volume\u0027s connect_volume method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c52f0a8b2432de858f5592c298d74afc9984d1b4"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-17 16:32:01.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/53310a45cb36e54e61e0e48e821e6cc3df93fe2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/53310a45cb36e54e61e0e48e821e6cc3df93fe2f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c52f0a8b2432de858f5592c298d74afc9984d1b4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"997484eff9724c42e5f27ae19d8902fcb42c416c":{"kind":"REWORK","_number":9,"created":"2014-10-08 07:16:28.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/9"}}},"commit":{"parents":[{"commit":"f50c8dda2a9edaa75b9f13ffe7bc4e6a89c15d5f","subject":"Merge \"Enhance  V2 disk_config extension Unit Test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f50c8dda2a9edaa75b9f13ffe7bc4e6a89c15d5f"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-10-08 07:16:05.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\n5) Tests were edited accordingly\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/997484eff9724c42e5f27ae19d8902fcb42c416c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/997484eff9724c42e5f27ae19d8902fcb42c416c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f50c8dda2a9edaa75b9f13ffe7bc4e6a89c15d5f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0f6a84626420df7d02517541593dcd55d62fdaac":{"kind":"REWORK","_number":10,"created":"2014-12-01 09:57:49.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/10"}}},"commit":{"parents":[{"commit":"59acb16fe3fcba6342d6a02ca69df9b64eb3332e","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59acb16fe3fcba6342d6a02ca69df9b64eb3332e"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-12-01 08:58:25.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\n5) Tests were edited accordingly\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f6a84626420df7d02517541593dcd55d62fdaac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f6a84626420df7d02517541593dcd55d62fdaac"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"59acb16fe3fcba6342d6a02ca69df9b64eb3332e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"79034f3daddc6d162490450944f0fdb7f8dc9f66":{"kind":"REWORK","_number":11,"created":"2014-12-01 14:45:52.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/11"}}},"commit":{"parents":[{"commit":"8425fbc01c958591ba8eb33514fa78b788f73a86","subject":"Merge \"Implement \u0027personality\u0027 plugin for V2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8425fbc01c958591ba8eb33514fa78b788f73a86"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-12-01 13:47:50.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\n1) All single \u0027insert\u0027 operations were changed by\nsingle \u0027multiline insert\u0027 operation\n\n2) New paramater \u0027want_result\u0027 was added to this function,\nbecause one extra \u0027select\u0027 operation is not always needed\n\n3) Explicit flush() was removed\n\n4) Using code from oslo.db provides\ngetting error value from DBDuplicateEntry exception\n\n5) Tests were edited accordingly\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/79034f3daddc6d162490450944f0fdb7f8dc9f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/79034f3daddc6d162490450944f0fdb7f8dc9f66"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8425fbc01c958591ba8eb33514fa78b788f73a86","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"44f5c16f10d6b646c2e197b3b9fa7eb9ec184935":{"kind":"REWORK","_number":12,"created":"2014-12-01 17:10:08.000000000","uploader":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"ref":"refs/changes/75/118675/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/118675/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/118675/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/118675/12"}}},"commit":{"parents":[{"commit":"e8b085293cf525d16570afa01e4a07da84e6fe8b","subject":"Merge \"Enable pep8 on ./tools directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e8b085293cf525d16570afa01e4a07da84e6fe8b"}]}],"author":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-09-04 15:51:27.000000000","tz":240},"committer":{"name":"pkholkin","email":"pkholkin@mirantis.com","date":"2014-12-01 16:12:06.000000000","tz":240},"subject":"Optimize \u0027floating_ip_bulk_create\u0027 function","message":"Optimize \u0027floating_ip_bulk_create\u0027 function\n\nThe existing floating_ip_bulk_create() DB API method used a for loop and\na session flush operation for each IP created. This patch consolidates\nthose many operations into a single INSERT statement, which will\nperform better than multiple round-trips to the SQL database. We further\noptimize the function by making a return of the changed rows optional.\n\nChange-Id: I5cf075f38696110d2890cfa1ea488ca459d791a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44f5c16f10d6b646c2e197b3b9fa7eb9ec184935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44f5c16f10d6b646c2e197b3b9fa7eb9ec184935"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e8b085293cf525d16570afa01e4a07da84e6fe8b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
