)]}'
{"id":"openstack%2Fnova~121965","triplet_id":"openstack%2Fnova~master~I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f","project":"openstack/nova","branch":"master","topic":"bug/1359617.2","hashtags":[],"change_id":"I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f","subject":"Make separate calls to libvirt volume","status":"MERGED","created":"2014-09-16 19:50:44.000000000","updated":"2014-10-29 18:02:26.000000000","submitted":"2014-10-29 18:02:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0bea86bdbb6bf5606e8b004126b6ee873921a1ef","_number":121965,"virtual_id_number":121965,"owner":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2014-10-08 23:51:30.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2014-10-07 20:03:34.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":2,"date":"2014-10-29 18:02:23.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"date":"2014-10-29 12:56:13.000000000","_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":1,"date":"2014-10-07 19:17:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-10-08 10:19:28.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":2,"date":"2014-10-29 03:50:19.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":1,"date":"2014-10-29 03:50:19.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-09-18 05:32:06.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2014-09-29 15:27:12.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"},{"updated":"2014-09-30 19:16:12.000000000","updated_by":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"reviewer":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"state":"REVIEWER"},{"updated":"2014-09-30 20:13:58.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-07 19:17:38.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-07 20:03:34.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-08 10:19:28.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2014-10-08 23:51:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-29 03:50:19.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2014-10-29 18:02:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b1530f4fd4fe021d126d54880f10e6d1567e0c58","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-16 19:50:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58d9628fa277fb1592615af0acb1d63723a29a1c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-16 20:13:40.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/849d5a369d1d470097fcec7712cfab5f/131007_devstack_export.log : SUCCESS in 14m 36s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d48df308cad04ffca5dda218ef395841/user_001.log : SUCCESS in 20m 50s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0da7a29f5d474c44bb3d646b77097c1a/131007_devstack_export.log : SUCCESS in 15m 45s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_user_001:th-percona/9f081e09f0b2411ba3b5373cf717c8bc/user_001.log : SUCCESS in 21m 10s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f07b10d703b84c90b6f775d49b6776b5/user_002.log : SUCCESS in 21m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_user_002:th-percona/1c2592f25f394414a5bc6ddb5dff0584/user_002.log : SUCCESS in 18m 34s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa2818efc2a2b48a476b7dfbba9effacbfa4ed66","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-16 21:01:01.000000000","message":"Patch Set 1:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/1/26385/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"1e946105ce4ea70ef4b24269dbafcf15faaaac2d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-16 21:04:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/1/ : Looks good to me in 1h 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f9543a9e8ee26d01bb90e4c8a1d56c1e2c8503ea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 21:53:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/1/check/gate-nova-pep8/fdaa11f : SUCCESS in 12m 30s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/1/check/gate-nova-docs/e0aa8bf/doc/build/html/ : SUCCESS in 24m 52s\n- gate-nova-python26 http://logs.openstack.org/65/121965/1/check/gate-nova-python26/e1e63cd : SUCCESS in 26m 05s\n- gate-nova-python27 http://logs.openstack.org/65/121965/1/check/gate-nova-python27/06fe238 : SUCCESS in 20m 34s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-full/170f1ab : FAILURE in 47m 58s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-postgres-full/5769a78 : FAILURE in 1h 10m 35s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-neutron-full/e342279 : FAILURE in 49m 48s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-neutron-heat-slow/d5024c4 : SUCCESS in 24m 47s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/1/check/check-grenade-dsvm/3bb4394 : FAILURE in 39m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/1/check/gate-tempest-dsvm-large-ops/e3f2a30 : SUCCESS in 24m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/1/check/gate-tempest-dsvm-neutron-large-ops/bacb747 : SUCCESS in 23m 51s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/1/check/check-devstack-dsvm-cells/67bcb2d : SUCCESS in 22m 28s\n- gate-nova-pylint http://logs.openstack.org/65/121965/1/check/gate-nova-pylint/df930a9 : SUCCESS in 22m 09s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/1/check/check-grenade-dsvm-partial-ncpu/e9fb783 : FAILURE in 32m 07s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-ironic-pxe_ssh-nv/dd857b8 : SUCCESS in 37m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ecb7c6e7aad2956ba64fdd55d4277c205cd0df4","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-16 23:11:34.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"c47cf5c21d16423ba0a332d443dddae995aab7b2","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-16 23:38:14.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b61c0fee872e4b4ab5d2b172568378a8/131007_devstack_export.log : SUCCESS in 19m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f6836f7e723c43008c541128b363ae60/user_001.log : SUCCESS in 23m 20s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b22cb3d3d96e4030a71c37dc80ef3e7b/131007_devstack_export.log : SUCCESS in 18m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_user_001:th-percona/70c6156ccd9442469a00092076609cc0/user_001.log : SUCCESS in 26m 00s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/add56710f48a4f1081f9bd30386d2212/user_002.log : SUCCESS in 18m 11s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_user_002:th-percona/a6b76171b56744a3bd1e299bb7e5104e/user_002.log : SUCCESS in 18m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"6fb0bd0f44c80b952e6e614b7935432f41e41982","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-17 01:08:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/1/ : Looks good to me in 1h 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2dab3bcd0afbadda340444df77c6d8294837630e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-17 01:25:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/1/check/gate-nova-pep8/ee52897 : SUCCESS in 12m 51s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/1/check/gate-nova-docs/b9d8338/doc/build/html/ : SUCCESS in 16m 46s\n- gate-nova-python26 http://logs.openstack.org/65/121965/1/check/gate-nova-python26/59ffd54 : FAILURE in 1h 10m 21s\n- gate-nova-python27 http://logs.openstack.org/65/121965/1/check/gate-nova-python27/daa5db2 : SUCCESS in 13m 47s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-full/c699134 : FAILURE in 57m 02s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-postgres-full/0df8643 : FAILURE in 1h 01m 37s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-neutron-full/424ea8f : FAILURE in 1h 07m 33s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-neutron-heat-slow/ababf7b : SUCCESS in 25m 26s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/1/check/check-grenade-dsvm/c8393ed : FAILURE in 39m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/1/check/gate-tempest-dsvm-large-ops/3e9e336 : SUCCESS in 20m 04s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/1/check/gate-tempest-dsvm-neutron-large-ops/0383384 : SUCCESS in 27m 33s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/1/check/check-devstack-dsvm-cells/5eab52c : SUCCESS in 14m 15s\n- gate-nova-pylint http://logs.openstack.org/65/121965/1/check/gate-nova-pylint/7d0e605 : SUCCESS in 22m 56s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/1/check/check-grenade-dsvm-partial-ncpu/4bdb177 : FAILURE in 25m 43s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-ironic-pxe_ssh-nv/c69a980 : SUCCESS in 35m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3deeb646ca93982bbbf16482e56e1033855dd2e","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-17 02:32:00.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"100613324a898dfd1d2567d2c76c5b301ca36eff","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-17 02:59:34.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/92872ca75e97413ba9e19db1f3ac1db4/131007_devstack_export.log : SUCCESS in 15m 22s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/032f799daa5b4183ab19fd302fc4796b/user_001.log : SUCCESS in 21m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4bd20205920145b090830dbed347471c/131007_devstack_export.log : SUCCESS in 17m 57s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_user_001:th-percona/fc8c97c892a74f9e8da28ec3265a0316/user_001.log : SUCCESS in 26m 45s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ca3001dab2064ecbb1aad2333b4c7d74/user_002.log : SUCCESS in 18m 06s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/1/check/real-db-upgrade_nova_percona_user_002:th-percona/ba9fe5ec9efd4fc292e9e5836f3a45ec/user_002.log : SUCCESS in 18m 06s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"359f75c4f8a4ed8cba35b355c60a054f2d6c58e3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-17 03:44:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/1/ : Looks good to me in 1h 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"481f97b04c756b55f3b27856897869b02215afb5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-17 04:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/1/check/gate-nova-pep8/891d1b8 : SUCCESS in 8m 54s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/1/check/gate-nova-docs/49b64e8/doc/build/html/ : SUCCESS in 12m 20s\n- gate-nova-python26 http://logs.openstack.org/65/121965/1/check/gate-nova-python26/26c5432 : SUCCESS in 18m 32s\n- gate-nova-python27 http://logs.openstack.org/65/121965/1/check/gate-nova-python27/7442c6e : SUCCESS in 16m 40s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-full/30d2a4d : FAILURE in 53m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-postgres-full/20e0c07 : FAILURE in 1h 18m 18s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-neutron-full/8ead61b : FAILURE in 1h 02m 27s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-neutron-heat-slow/c404b07 : SUCCESS in 19m 50s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/1/check/check-grenade-dsvm/910e784 : FAILURE in 38m 32s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/1/check/gate-tempest-dsvm-large-ops/af8c982 : SUCCESS in 22m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/1/check/gate-tempest-dsvm-neutron-large-ops/762024f : SUCCESS in 18m 18s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/1/check/check-devstack-dsvm-cells/6dd7235 : SUCCESS in 14m 27s\n- gate-nova-pylint http://logs.openstack.org/65/121965/1/check/gate-nova-pylint/3d187c3 : SUCCESS in 22m 06s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/1/check/check-grenade-dsvm-partial-ncpu/f452883 : SUCCESS in 37m 19s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/1/check/check-tempest-dsvm-ironic-pxe_ssh-nv/56b14d8 : SUCCESS in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b7b362b983387622a274fe646f99a6f885363ad","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-09-17 11:23:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b14850d6ff8991f33cf9dd53e236024b075f1373","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-17 16:45:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"6568f7657e569c3f0a9efb5a9849364f4904a3a6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-17 17:53:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/2/ : Looks good to me in 1h 07m 12s","accounts_in_message":[],"_revision_number":2},{"id":"aab3f10de03581b9ebe8ee491e58edda4facd6f1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-17 18:03:40.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/2/26540/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"0e9fb89c870784da006acb2d09effcbf19ed5331","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-17 19:07:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/2/check/gate-nova-pep8/2f9b7ae : SUCCESS in 13m 42s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/2/check/gate-nova-docs/fb20ae8/doc/build/html/ : SUCCESS in 18m 33s\n- gate-nova-python26 http://logs.openstack.org/65/121965/2/check/gate-nova-python26/bd9e52c : SUCCESS in 24m 08s\n- gate-nova-python27 http://logs.openstack.org/65/121965/2/check/gate-nova-python27/271063e : SUCCESS in 14m 00s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/2/check/check-tempest-dsvm-full/e44d074 : FAILURE in 54m 50s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/2/check/check-tempest-dsvm-postgres-full/b50d705 : FAILURE in 1h 01m 46s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/2/check/check-tempest-dsvm-neutron-full/d5545ae : FAILURE in 47m 35s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/2/check/check-tempest-dsvm-neutron-heat-slow/a036593 : SUCCESS in 19m 55s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/2/check/check-grenade-dsvm/7063349 : FAILURE in 34m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/2/check/gate-tempest-dsvm-large-ops/72c8781 : SUCCESS in 19m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/2/check/gate-tempest-dsvm-neutron-large-ops/95e1674 : SUCCESS in 18m 17s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/2/check/check-devstack-dsvm-cells/01d363e : SUCCESS in 13m 29s\n- gate-nova-pylint http://logs.openstack.org/65/121965/2/check/gate-nova-pylint/47a0552 : SUCCESS in 15m 50s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/2/check/check-grenade-dsvm-partial-ncpu/7dd12e6 : SUCCESS in 45m 26s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/2/check/check-tempest-dsvm-ironic-pxe_ssh-nv/39beee2 : SUCCESS in 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92a66a1a3c65559ca4fefc6d9b6d50a08be4d3be","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 01:20:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d1bd1ab9e1468ee368c393b0c311cb68b7d97ab5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-18 01:24:01.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/121965/3/ : Test run failed. in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1dcbc9fd73f66d080b6b1157b76834b0489e6ac1","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 02:26:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9e3f0b05a0940676fef33824da7bdc00d17faa3b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-18 02:29:54.000000000","message":"Patch Set 4:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/121965/4/ : Test run failed. in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"4f5a78dcc1821a2475bc090b35791947aa13450c","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 02:55:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4cb4db60281ff89f9b8243f43240e96931a0f117","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-18 02:59:22.000000000","message":"Patch Set 5:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/121965/5/ : Test run failed. in 3m 06s","accounts_in_message":[],"_revision_number":5},{"id":"4bf92bc8754ba9351725ec4387d5a318b6c48c7e","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 04:06:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bd34e951523375fd0cadb1953e699ba4bf1974b2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-18 04:08:40.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/5/26590/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"ed81643d93bc001c12bc8b8968555b85b1daee8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-18 04:23:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/5/check/gate-nova-pep8/8e2354b : SUCCESS in 11m 33s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/5/check/gate-nova-docs/ccb9fb9/doc/build/html/ : SUCCESS in 15m 41s\n- gate-nova-python26 http://logs.openstack.org/65/121965/5/check/gate-nova-python26/b23eb1a : SUCCESS in 25m 04s\n- gate-nova-python27 http://logs.openstack.org/65/121965/5/check/gate-nova-python27/398b4ed : SUCCESS in 15m 14s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/5/check/check-tempest-dsvm-full/9eedb83 : FAILURE in 57m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/5/check/check-tempest-dsvm-postgres-full/0fd4c6f : FAILURE in 1h 20m 02s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/5/check/check-tempest-dsvm-neutron-full/fcfd52d : SUCCESS in 1h 04m 23s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/5/check/check-tempest-dsvm-neutron-heat-slow/667f4b9 : SUCCESS in 24m 31s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/5/check/check-grenade-dsvm/4e26210 : SUCCESS in 46m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/5/check/gate-tempest-dsvm-large-ops/48cc268 : SUCCESS in 19m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/5/check/gate-tempest-dsvm-neutron-large-ops/caf2a28 : SUCCESS in 23m 28s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/5/check/check-devstack-dsvm-cells/56e5089 : SUCCESS in 17m 06s\n- gate-nova-pylint http://logs.openstack.org/65/121965/5/check/gate-nova-pylint/cd1a7d2 : SUCCESS in 15m 08s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/5/check/check-grenade-dsvm-partial-ncpu/b2f2732 : SUCCESS in 45m 19s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/5/check/check-tempest-dsvm-ironic-pxe_ssh-nv/cf16fc8 : SUCCESS in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dd71ca5cb160da4414d101fc43790e2904e2b1ae","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 04:39:09.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"ee24471b6c45ab55821fa9d164bdceb051bbd2fe","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 04:50:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8dd42ed0871eb94055f39c85cd3e7bdb6df9dcd1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-18 04:54:07.000000000","message":"Patch Set 6:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/121965/6/ : Test run failed. in 3m 03s","accounts_in_message":[],"_revision_number":6},{"id":"761d1532cc3bd4711f9c625dabad02cba2e047e1","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2014-09-18 05:32:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fab78fd67ebca607dd3b4a98b7b8e2f11ecd320b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-18 06:01:02.000000000","message":"Patch Set 6: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/6/26600/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"1806831b19622ba20b251adf1b7a4a2a1cdf6bfa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-18 06:19:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/6/check/gate-nova-pep8/c579af1 : SUCCESS in 8m 40s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/6/check/gate-nova-docs/18fac50/doc/build/html/ : SUCCESS in 13m 04s\n- gate-nova-python26 http://logs.openstack.org/65/121965/6/check/gate-nova-python26/ea83ba9 : SUCCESS in 22m 31s\n- gate-nova-python27 http://logs.openstack.org/65/121965/6/check/gate-nova-python27/9411d2b : SUCCESS in 13m 23s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/6/check/check-tempest-dsvm-full/e532140 : SUCCESS in 45m 24s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/6/check/check-tempest-dsvm-postgres-full/3b83663 : SUCCESS in 1h 27m 55s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/6/check/check-tempest-dsvm-neutron-full/1b97f5c : SUCCESS in 1h 04m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/6/check/check-tempest-dsvm-neutron-heat-slow/eca0664 : SUCCESS in 23m 50s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/6/check/check-grenade-dsvm/e272504 : SUCCESS in 44m 42s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/6/check/gate-tempest-dsvm-large-ops/7d346bb : SUCCESS in 24m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/6/check/gate-tempest-dsvm-neutron-large-ops/0932040 : SUCCESS in 23m 32s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/6/check/check-devstack-dsvm-cells/0304f9e : SUCCESS in 17m 03s\n- gate-nova-pylint http://logs.openstack.org/65/121965/6/check/gate-nova-pylint/40e823b : SUCCESS in 15m 02s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/6/check/check-grenade-dsvm-partial-ncpu/19ff5fa : SUCCESS in 44m 34s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/6/check/check-tempest-dsvm-ironic-pxe_ssh-nv/9b30961 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1ebc3d4bfd8d8689258ede549692ca4a79088082","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-18 11:00:03.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3baf0f869d0f49d1a0c28cde64dbfbf3/131007_devstack_export.log : SUCCESS in 16m 09s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/87612881593945e788fd7730312dfe76/user_001.log : SUCCESS in 25m 15s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/15c0453e354440b2a1bee2a5cc9e4450/131007_devstack_export.log : SUCCESS in 15m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/6/check/real-db-upgrade_nova_percona_user_001:th-percona/95ec467f6ee944c4845f752cad95e771/user_001.log : SUCCESS in 26m 00s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c25b932dc80400fbcdbf91cc70a088f/user_002.log : SUCCESS in 18m 12s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/6/check/real-db-upgrade_nova_percona_user_002:th-percona/614c3b60beeb426fafeda737bf078689/user_002.log : SUCCESS in 22m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"aff12fe50fb676d5ea3ead2a132ac50fc35b7ecd","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-09-18 15:46:00.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nOne question about updating the connection info after get_config!","accounts_in_message":[],"_revision_number":6},{"id":"e554e6e933cbbcb186c7a6d3be3fd7ed320e551a","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 17:40:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9761ec2184112dbeb9214e953b35c2984f122f72","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 21:06:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4efbee32682fcdf0b900fc3439346b30cfd4fcb9","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 21:11:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"318dbe4f1e5a9839b0e63dbca28ead6683077bd7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-18 21:18:41.000000000","message":"Patch Set 8:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/8/26729/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":8},{"id":"45908463b50e399ec95cf2f197708d405cf8851d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-18 21:36:32.000000000","message":"Patch Set 8: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/02c4a2e0b6e74fefbd554c07cada5390/131007_devstack_export.log : SUCCESS in 15m 37s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3f632a5ad0394baa9f5a20b247978762/user_001.log : SUCCESS in 24m 33s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3e3fea677d1d46ed95b61c871eafd6c4/131007_devstack_export.log : SUCCESS in 14m 54s\n- real-db-upgrade_nova_percona_user_001:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1972421.slave.openstack.org-4/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 3m 27s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f3fe04c0fbda4a5db6ffa24658b7637e/user_002.log : SUCCESS in 17m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://localhost : Exception: [Errno 2] No such file or directory: \u0027/var/lib/turbo-hipster/git/real_db_upgrade-th-percona-rax-syd-1972414.slave.openstack.org-5/openstack/nova/nova/db/sqlalchemy/migrate_repo/versions\u0027 in 2m 36s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"742e20a960c2f6931edb150152f5ebc7fa2397c4","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-18 21:45:32.000000000","message":"Patch Set 8:\n\nrecheck migrations","accounts_in_message":[],"_revision_number":8},{"id":"021e33edaab0d4c5e5878788acb48010c5c1a1ad","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-18 22:13:33.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/c34b35322475483db3ae9f3eba2b7ec6/131007_devstack_export.log : SUCCESS in 15m 16s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/73f2ca9c78ac4173bf76de57d8963266/user_001.log : SUCCESS in 27m 27s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/22a2507e5121400ebffba1c7fd9e4519/131007_devstack_export.log : SUCCESS in 15m 31s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_percona_user_001:th-percona/7a33c6435f78442aa8d92a37df2079a2/user_001.log : SUCCESS in 23m 22s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0cf196f71ce94e2b8a8374e63deb536d/user_002.log : SUCCESS in 18m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/8/check/real-db-upgrade_nova_percona_user_002:th-percona/371d73ea73ae4313bdb576d0f74a405d/user_002.log : SUCCESS in 21m 39s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"f1ed0554d0cd85891fd6d70cce0c3c9f88604f67","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-19 00:17:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/8/ : Looks good to me in 1h 19m 33s","accounts_in_message":[],"_revision_number":8},{"id":"538bf2a8398a957cebc153e7e08485fd9471c337","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-19 03:58:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/8/check/gate-nova-pep8/a1f0cc0 : SUCCESS in 10m 43s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/8/check/gate-nova-docs/e745b16/doc/build/html/ : SUCCESS in 12m 20s\n- gate-nova-python26 http://logs.openstack.org/65/121965/8/check/gate-nova-python26/3348bcb : SUCCESS in 16m 04s\n- gate-nova-python27 http://logs.openstack.org/65/121965/8/check/gate-nova-python27/6a830c7 : SUCCESS in 21m 35s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/8/check/check-tempest-dsvm-full/5ec81ee : SUCCESS in 47m 22s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/8/check/check-tempest-dsvm-postgres-full/03655a9 : SUCCESS in 1h 20m 00s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/8/check/check-tempest-dsvm-neutron-full/c0ce1aa : SUCCESS in 55m 31s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/8/check/check-tempest-dsvm-neutron-heat-slow/d93f65d : SUCCESS in 18m 59s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/8/check/check-grenade-dsvm/549dff7 : SUCCESS in 39m 00s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/8/check/gate-tempest-dsvm-large-ops/d370ef1 : SUCCESS in 25m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/8/check/gate-tempest-dsvm-neutron-large-ops/94d626e : SUCCESS in 23m 08s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/8/check/check-devstack-dsvm-cells/954452d : SUCCESS in 17m 14s\n- gate-nova-pylint http://logs.openstack.org/65/121965/8/check/gate-nova-pylint/4202ce8 : SUCCESS in 21m 42s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/8/check/check-grenade-dsvm-partial-ncpu/d788956 : SUCCESS in 44m 03s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/8/check/check-tempest-dsvm-ironic-pxe_ssh-nv/55c6f7a : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8543f7328cb8cea768398721c655c5cb945a874f","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-09-19 10:21:59.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)\n\nLong analysis inline","accounts_in_message":[],"_revision_number":8},{"id":"6e373979307a2f3c7e770fa9c9c613f01cc8681c","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-19 21:15:27.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bbbc74014fedbffe0c42f2f47664466ca9da4ebf","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-22 04:54:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"065a6e23b8a2f7f95764b416bb4d0d67f71cf311","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-22 05:22:31.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3d17920262fa486085f38533a372a8c8/131007_devstack_export.log : SUCCESS in 15m 36s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a7774e58a64c491aa6e2ff8e6ec6be5a/user_001.log : SUCCESS in 22m 05s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2c9d58c0a18a48b393f35cade83f9bfb/131007_devstack_export.log : SUCCESS in 16m 05s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/9/check/real-db-upgrade_nova_percona_user_001:th-percona/1f79484b50af413c958b4f2b9244c0dc/user_001.log : SUCCESS in 22m 00s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5ad77f2a463f460f8f2c2b70ff23be08/user_002.log : SUCCESS in 26m 43s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/9/check/real-db-upgrade_nova_percona_user_002:th-percona/12bac1ed03cf4b459d354636d3cf895d/user_002.log : SUCCESS in 18m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"e285a210c86df8cc771e4f38bf211577dc10743f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-22 05:58:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/9/ : Looks good to me in 1h 02m 37s","accounts_in_message":[],"_revision_number":9},{"id":"cf380b7a041b3c9ad8ab1a662e7a8707bb57e531","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-22 06:12:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/9/check/gate-nova-pep8/6ddd988 : SUCCESS in 12m 05s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/9/check/gate-nova-docs/fa8a25c/doc/build/html/ : SUCCESS in 15m 52s\n- gate-nova-python26 http://logs.openstack.org/65/121965/9/check/gate-nova-python26/052c23f : SUCCESS in 23m 43s\n- gate-nova-python27 http://logs.openstack.org/65/121965/9/check/gate-nova-python27/526eb33 : SUCCESS in 16m 52s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/9/check/check-tempest-dsvm-full/c6ea821 : SUCCESS in 58m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/9/check/check-tempest-dsvm-postgres-full/580092d : SUCCESS in 50m 46s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/9/check/check-tempest-dsvm-neutron-full/87454d0 : SUCCESS in 1h 10m 39s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/9/check/check-tempest-dsvm-neutron-heat-slow/75b2ccb : SUCCESS in 25m 09s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/9/check/check-grenade-dsvm/f6d4c14 : SUCCESS in 48m 48s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/9/check/gate-tempest-dsvm-large-ops/9d9be7b : SUCCESS in 24m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/9/check/gate-tempest-dsvm-neutron-large-ops/5d91e2b : SUCCESS in 22m 37s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/9/check/check-devstack-dsvm-cells/cea902f : SUCCESS in 14m 46s\n- gate-nova-pylint http://logs.openstack.org/65/121965/9/check/gate-nova-pylint/cf947e6 : SUCCESS in 15m 35s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/9/check/check-grenade-dsvm-partial-ncpu/4e43eac : SUCCESS in 46m 36s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/9/check/check-tempest-dsvm-ironic-pxe_ssh-nv/b38d9fc : FAILURE in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4942ba918bc19ab3e9af55f0edde6079331f0238","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-09-29 14:04:04.000000000","message":"Patch Set 9:\n\nBuild failed. Logs: http://208.91.1.172/logs/121965/9 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":9},{"id":"f75d84502fbca92abc01f10e704c9c03e95a3597","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-09-29 15:27:12.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nsome nits inline","accounts_in_message":[],"_revision_number":9},{"id":"d72d0f56f32bd294a9c783353648442a821b18ea","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-29 15:45:35.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4355fe2a071b1a83297cd9701e93a4b92b96e240","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 01:37:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"edda855874fd339b54e91f82848d45a818e7c10f","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 01:39:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"de533b667e63a1dbd65fff9244fb6f9e9ce5a16a","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 01:59:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7f1c2c5fa5058402d972986ca677ce726f4981a3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-30 02:04:23.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e57060ad26a440fc92f7deb0365f4d7b/131007_devstack_export.log : SUCCESS in 15m 41s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c8846b51a4034b95a7c13d2a8468a1e8/user_001.log : SUCCESS in 24m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3c4e762f9437439e917b910330bfea35/131007_devstack_export.log : SUCCESS in 15m 07s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_user_001:th-percona/bd23675e6bc84ac9a339c43dc5f8359b/user_001.log : SUCCESS in 22m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/bccbe8b8f6b24617beabed3141ba4db2/user_002.log : SUCCESS in 18m 31s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_user_002:th-percona/36856f4e305c49e891d1927867a03392/user_002.log : SUCCESS in 17m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"7c1ca6d8479ca2de475410ddf7aa1b8866715bd3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-30 02:46:18.000000000","message":"Patch Set 11:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/11/27043/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":11},{"id":"753a46c0454fae6ba1f3e53ce08b935a69c38f56","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-30 02:52:29.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/11/check/gate-nova-pep8/2bf1af9 : SUCCESS in 9m 02s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/11/check/gate-nova-docs/d9d6d32/doc/build/html/ : SUCCESS in 12m 33s\n- gate-nova-python26 http://logs.openstack.org/65/121965/11/check/gate-nova-python26/88b6b5b : SUCCESS in 16m 36s\n- gate-nova-python27 http://logs.openstack.org/65/121965/11/check/gate-nova-python27/029f722 : SUCCESS in 20m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-full/2e596fb : FAILURE in 1h 06m 10s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-postgres-full/6f1d921 : FAILURE in 55m 06s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-neutron-full/5a95543 : FAILURE in 1h 12m 05s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-neutron-heat-slow/4e7dba2 : SUCCESS in 25m 56s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/11/check/check-grenade-dsvm/5f5ca4f : FAILURE in 56m 35s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/11/check/gate-tempest-dsvm-large-ops/54f67bb : SUCCESS in 18m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/11/check/gate-tempest-dsvm-neutron-large-ops/4bfeeea : SUCCESS in 19m 09s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/11/check/check-devstack-dsvm-cells/681879a : SUCCESS in 14m 38s\n- gate-nova-pylint http://logs.openstack.org/65/121965/11/check/gate-nova-pylint/94ee4dd : SUCCESS in 15m 11s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/11/check/check-grenade-dsvm-partial-ncpu/28db1a8 : SUCCESS in 53m 54s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-ironic-pxe_ssh-nv/976cd29 : SUCCESS in 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c67373113846d327ccee72ae99ad4597f5c10355","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 03:04:05.000000000","message":"Patch Set 11:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":11},{"id":"93a0dffa11f13273cd7b4f4aa2c5081206b853a3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-30 03:10:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/11/ : Looks good to me in 1h 30m 58s","accounts_in_message":[],"_revision_number":11},{"id":"8d61edc0967c1f2f1deb590bfc4284367a9f7844","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-30 03:28:32.000000000","message":"Patch Set 11:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8e7a2711d35d4841a1b182cc273072c1/131007_devstack_export.log : SUCCESS in 18m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6a8065591e18427497137414cd5bfe31/user_001.log : SUCCESS in 23m 51s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c96f3caf860a4c07846a5f8c23d7152a/131007_devstack_export.log : SUCCESS in 15m 03s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_user_001:th-percona/833a1133bc3a466fa6f72e6ba8c531ae/user_001.log : SUCCESS in 22m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/bda9c8c185d045579d305bce783c4651/user_002.log : SUCCESS in 21m 14s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_user_002:th-percona/32c8e510aa334e26ad403d5afef4e13f/user_002.log : SUCCESS in 18m 24s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"38f0053834c5d31327a41ed58ee2c8ac183b6ee7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-30 04:02:17.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/11/check/gate-nova-pep8/29c4015 : SUCCESS in 13m 03s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/11/check/gate-nova-docs/9d5d0bc/doc/build/html/ : SUCCESS in 21m 28s\n- gate-nova-python26 http://logs.openstack.org/65/121965/11/check/gate-nova-python26/e23c19e : SUCCESS in 16m 47s\n- gate-nova-python27 http://logs.openstack.org/65/121965/11/check/gate-nova-python27/53dbd7c : SUCCESS in 20m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-full/6cc0e96 : FAILURE in 55m 37s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-postgres-full/e16d76e : FAILURE in 46m 57s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-neutron-full/3a726ff : FAILURE in 57m 38s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-neutron-heat-slow/c4241d7 : SUCCESS in 23m 07s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/11/check/check-grenade-dsvm/6ad8955 : FAILURE in 53m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/11/check/gate-tempest-dsvm-large-ops/b31a16b : SUCCESS in 23m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/11/check/gate-tempest-dsvm-neutron-large-ops/9a921fb : SUCCESS in 20m 57s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/11/check/check-devstack-dsvm-cells/ff4466f : SUCCESS in 18m 31s\n- gate-nova-pylint http://logs.openstack.org/65/121965/11/check/gate-nova-pylint/a006b18 : SUCCESS in 14m 29s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/11/check/check-grenade-dsvm-partial-ncpu/44b34df : SUCCESS in 46m 34s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-ironic-pxe_ssh-nv/17ac9e4 : SUCCESS in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d9c7a2b0abab401698f5e3f45fd3b236f57cbb9b","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 04:04:56.000000000","message":"Patch Set 11:\n\nrecheck bug 1348204","accounts_in_message":[],"_revision_number":11},{"id":"64220cd2b37d4663424b9c88c35556a68e42810e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-30 04:29:03.000000000","message":"Patch Set 11:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5d3977a2b1fa43bab92d05272f7e43ad/131007_devstack_export.log : SUCCESS in 16m 53s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/be18554b87364030ad4af8f386049401/user_001.log : SUCCESS in 21m 55s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7c4df2517ee14ea880012cceb16ca7c5/131007_devstack_export.log : SUCCESS in 15m 57s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_user_001:th-percona/d3b93113d12c4246809a6f88bfda9066/user_001.log : SUCCESS in 23m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c4d6a160cf2a4143a7f60a7c7d20816c/user_002.log : SUCCESS in 18m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/11/check/real-db-upgrade_nova_percona_user_002:th-percona/e7af90ae8ec442c7973b333bc75e8e25/user_002.log : SUCCESS in 17m 35s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"01dc2df4763640c9b7753187e498ad4da10e1aae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-30 05:11:30.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/11/check/gate-nova-pep8/3e25268 : SUCCESS in 12m 16s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/11/check/gate-nova-docs/ca84db8/doc/build/html/ : SUCCESS in 24m 18s\n- gate-nova-python26 http://logs.openstack.org/65/121965/11/check/gate-nova-python26/c2797e9 : SUCCESS in 15m 16s\n- gate-nova-python27 http://logs.openstack.org/65/121965/11/check/gate-nova-python27/7536251 : SUCCESS in 13m 44s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-full/11a76aa : FAILURE in 56m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-postgres-full/6ba51bd : FAILURE in 59m 35s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-neutron-full/0358d8b : FAILURE in 1h 05m 54s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-neutron-heat-slow/4e54f2f : SUCCESS in 19m 57s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/11/check/check-grenade-dsvm/d7e87e5 : FAILURE in 1h 00m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/11/check/gate-tempest-dsvm-large-ops/b93872a : SUCCESS in 23m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/11/check/gate-tempest-dsvm-neutron-large-ops/e8efdd7 : SUCCESS in 25m 05s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/11/check/check-devstack-dsvm-cells/7011dfd : SUCCESS in 18m 10s\n- gate-nova-pylint http://logs.openstack.org/65/121965/11/check/gate-nova-pylint/b3a5581 : SUCCESS in 14m 51s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/11/check/check-grenade-dsvm-partial-ncpu/a5186dc : SUCCESS in 59m 58s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/11/check/check-tempest-dsvm-ironic-pxe_ssh-nv/c556c2c : SUCCESS in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c8a35bf6e98c39f8470ad604aee7f9161577bbfe","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-30 05:16:20.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/11/ : Looks good to me in 1h 11m 05s","accounts_in_message":[],"_revision_number":11},{"id":"e3f0de8b1cbd5b101ba68e3de9f241ee4c487d6c","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 15:25:00.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5a89e5fa9e2e57f73818a5f60bfd57aec0b95041","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 15:25:37.000000000","message":"Patch Set 12: Workflow-1\n\nTrying to figure out the tempest build failures.","accounts_in_message":[],"_revision_number":12},{"id":"9a298447b3996c4176e997fbbe83d4e1ddec6419","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-30 15:53:28.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/376425b8813445bf9b7c43dccc3dedaa/131007_devstack_export.log : SUCCESS in 15m 20s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/569429cf0e754daf9817739e1b02f378/user_001.log : SUCCESS in 21m 10s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/86170e500b394e0dae6c3c2ff0011b8f/131007_devstack_export.log : SUCCESS in 15m 54s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/12/check/real-db-upgrade_nova_percona_user_001:th-percona/e646992ad5284f12be0d263ce7ad498c/user_001.log : SUCCESS in 27m 38s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1ce894c94d484cf5aa758231fd1ba748/user_002.log : SUCCESS in 18m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/12/check/real-db-upgrade_nova_percona_user_002:th-percona/a4f1a47a675f49b6b9b74f340cfc013c/user_002.log : SUCCESS in 18m 35s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"a9f009636cb279b3141b026143b99bb21f378eaf","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-30 16:25:57.000000000","message":"Patch Set 12:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/121965/12/ : Test run failed. in 3m 41s","accounts_in_message":[],"_revision_number":12},{"id":"6df246b0d06485f011b4fd91b9ff811af8caaaf6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-30 16:33:13.000000000","message":"Patch Set 12:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/12/27119/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":12},{"id":"dbdd05929bf366717780be1883602afb649ae2d8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-30 16:33:42.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/12/check/gate-nova-pep8/68773df : SUCCESS in 13m 13s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/12/check/gate-nova-docs/f9c426d/doc/build/html/ : SUCCESS in 22m 44s\n- gate-nova-python26 http://logs.openstack.org/65/121965/12/check/gate-nova-python26/0aaafb7 : SUCCESS in 15m 49s\n- gate-nova-python27 http://logs.openstack.org/65/121965/12/check/gate-nova-python27/5396d3c : SUCCESS in 20m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/12/check/check-tempest-dsvm-full/db12b54 : SUCCESS in 57m 19s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/12/check/check-tempest-dsvm-postgres-full/32bce49 : SUCCESS in 58m 09s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/12/check/check-tempest-dsvm-neutron-full/1716f5f : SUCCESS in 1h 07m 35s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/12/check/check-tempest-dsvm-neutron-heat-slow/51d86e7 : SUCCESS in 22m 06s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/12/check/check-grenade-dsvm/603c039 : SUCCESS in 46m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/12/check/gate-tempest-dsvm-large-ops/2c1126b : SUCCESS in 18m 36s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/12/check/gate-tempest-dsvm-neutron-large-ops/9cbfb50 : SUCCESS in 18m 24s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/12/check/check-devstack-dsvm-cells/862903c : SUCCESS in 13m 37s\n- gate-nova-pylint http://logs.openstack.org/65/121965/12/check/gate-nova-pylint/3627c0c : SUCCESS in 15m 25s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/12/check/check-grenade-dsvm-partial-ncpu/92fb4c8 : SUCCESS in 45m 52s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/12/check/check-tempest-dsvm-ironic-pxe_ssh-nv/c056292 : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ff5bb69d5ea2b9cd166fb16ff6c0df14a67bb187","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 16:34:02.000000000","message":"Patch Set 12: -Workflow","accounts_in_message":[],"_revision_number":12},{"id":"8384b415fe287ee1f83e5568c9909f42417aba81","author":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"date":"2014-09-30 18:31:00.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)\n\nA minor nit","accounts_in_message":[],"_revision_number":12},{"id":"82491c894009d3ecd07d7e0a1f411057f9129dfd","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 18:32:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b6b5307242741378fd83efb6bf1a8692961dfc6c","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 19:01:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"329ef5b3c516df245439e7f1f1d4862eb31819cb","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-09-30 19:04:44.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8b93f80f68ebe1810514912ece0c28e637b86774","author":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"date":"2014-09-30 19:16:12.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"8eee401c9b0ae45672230a225010fb1c9e14a0a8","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-09-30 19:28:45.000000000","message":"Patch Set 13: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/13/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5f591e02fadf4e01b1f476dd066f2844/131007_devstack_export.log : SUCCESS in 16m 37s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/13/check/real-db-upgrade_nova_mysql_user_001:th-mysql/05ae9c3ec67d431ca6f72b7930716c03/user_001.log : SUCCESS in 21m 38s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/13/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b3b054afa86446f4b4f26697b92e1406/131007_devstack_export.log : SUCCESS in 15m 56s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/13/check/real-db-upgrade_nova_percona_user_001:th-percona/02522a5af14c4972819ba50a119be35b/user_001.log : SUCCESS in 23m 03s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/13/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e59b9d79da1d499c97e589c682b9e294/user_002.log : SUCCESS in 18m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/13/check/real-db-upgrade_nova_percona_user_002:th-percona/31a7a70941e543ddb38ad1227300b068/user_002.log : SUCCESS in 22m 24s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":13},{"id":"cca9b796e4d3d39bf7db9e77926627956712004a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-09-30 20:13:58.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/121965/13/ : Looks good to me in 1h 08m 46s","accounts_in_message":[],"_revision_number":13},{"id":"9eb5cce6ed2b9d7d9e66ba1a35b1d3b736b81c32","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-30 20:20:25.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/13/check/gate-nova-pep8/457e545 : SUCCESS in 11m 42s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/13/check/gate-nova-docs/2191c8d/doc/build/html/ : SUCCESS in 25m 23s\n- gate-nova-python26 http://logs.openstack.org/65/121965/13/check/gate-nova-python26/3c16b12 : SUCCESS in 24m 06s\n- gate-nova-python27 http://logs.openstack.org/65/121965/13/check/gate-nova-python27/b6ca9ee : SUCCESS in 20m 13s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/13/check/check-tempest-dsvm-full/8100f16 : SUCCESS in 57m 08s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/13/check/check-tempest-dsvm-postgres-full/ab3da3e : SUCCESS in 1h 00m 36s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/13/check/check-tempest-dsvm-neutron-full/14b86fb : SUCCESS in 1h 14m 29s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/13/check/check-tempest-dsvm-neutron-heat-slow/0370c3b : SUCCESS in 19m 16s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/13/check/check-grenade-dsvm/9c815aa : SUCCESS in 56m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/13/check/gate-tempest-dsvm-large-ops/0b9fbee : SUCCESS in 25m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/13/check/gate-tempest-dsvm-neutron-large-ops/41a22be : SUCCESS in 19m 22s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/13/check/check-devstack-dsvm-cells/4c463ca : SUCCESS in 17m 04s\n- gate-nova-pylint http://logs.openstack.org/65/121965/13/check/gate-nova-pylint/fbeb69c : SUCCESS in 15m 28s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/13/check/check-grenade-dsvm-partial-ncpu/2d86763 : SUCCESS in 53m 39s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/13/check/check-tempest-dsvm-ironic-pxe_ssh-nv/aeffb88 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b5a81ac6a15b73c7170ad998f0e9f62e2d4c4a17","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-09-30 21:00:25.000000000","message":"Patch Set 13:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/13/27133/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":13},{"id":"ce74e146935e7c7cc0b4dd93cb92af357cfe3180","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-07 16:30:14.000000000","message":"Patch Set 13: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"dd0fde9458c77902b578a2c1baa6cab3ae51f0ab","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-10-07 18:50:07.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"baea5a86713a6a319c9fa26cf8224e47293a4673","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-10-07 19:17:38.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f0537c6699d349dc98cec42048a3d67e/131007_devstack_export.log : SUCCESS in 17m 03s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f9a9e7ac20c848cf817e8cd2d96bcccc/user_001.log : SUCCESS in 26m 34s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d266fed912c447e38729e9302a0e01df/131007_devstack_export.log : SUCCESS in 16m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/14/check/real-db-upgrade_nova_percona_user_001:th-percona/745a88052b9c44119d83960b4e00133e/user_001.log : SUCCESS in 23m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/65/121965/14/check/real-db-upgrade_nova_mysql_user_002:th-mysql/92e994422cb049f9a82ce8903d354a71/user_002.log : SUCCESS in 19m 12s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/65/121965/14/check/real-db-upgrade_nova_percona_user_002:th-percona/a8e4e2d7d4aa46ef9139d071c4d024b1/user_002.log : SUCCESS in 19m 58s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"26d3cb9576a7c91b4220f6a1b5b4d39df62e46fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-07 19:58:46.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/14/check/gate-nova-pep8/0db0cbc : SUCCESS in 12m 27s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/14/check/gate-nova-docs/bed6f0a/doc/build/html/ : SUCCESS in 7m 23s\n- gate-nova-python26 http://logs.openstack.org/65/121965/14/check/gate-nova-python26/634f883 : SUCCESS in 16m 09s\n- gate-nova-python27 http://logs.openstack.org/65/121965/14/check/gate-nova-python27/e483c4e : SUCCESS in 18m 59s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-full/7953322 : SUCCESS in 57m 45s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-postgres-full/78c5b24 : SUCCESS in 56m 07s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-neutron-full/e4be86d : SUCCESS in 1h 07m 23s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-neutron-heat-slow/9e9453e : SUCCESS in 23m 42s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/14/check/check-grenade-dsvm/767a25a : SUCCESS in 51m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/14/check/gate-tempest-dsvm-large-ops/b899610 : SUCCESS in 20m 21s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/14/check/gate-tempest-dsvm-neutron-large-ops/e7591bd : SUCCESS in 23m 25s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/14/check/check-devstack-dsvm-cells/7f21acc : SUCCESS in 15m 57s\n- gate-nova-pylint http://logs.openstack.org/65/121965/14/check/gate-nova-pylint/ba2779f : SUCCESS in 14m 55s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/14/check/check-grenade-dsvm-partial-ncpu/bafe6ad : SUCCESS in 48m 19s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-ironic-pxe_ssh-nv/2491e0b : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"87468ca8950c29a25475e81eb5bb92e9bc906ccb","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-10-07 20:03:34.000000000","message":"Patch Set 14: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/65/121965/14/27753/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":14},{"id":"24efacfb8ca529ea5b5883079430d618c4a4649e","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-10-08 10:19:28.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"0d8c797635bd679a5783d777c4588195566b76ce","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-10-08 23:51:30.000000000","message":"Patch Set 14: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/121965/14","accounts_in_message":[],"_revision_number":14},{"id":"1e543e07725f5690105911a54d24bdea56a71906","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-10-29 03:50:19.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"8d5a5a5bb4796d817c7b9b790313c89ff964408a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 03:50:21.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"7e3499fd3f3783971ffc28485ab4da504de3b3c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 06:27:21.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/14/gate/gate-nova-docs/bd9280a/doc/build/html/ : SUCCESS in 8m 59s\n- gate-nova-pep8 http://logs.openstack.org/65/121965/14/gate/gate-nova-pep8/7513660 : SUCCESS in 10m 53s\n- gate-nova-python27 http://logs.openstack.org/65/121965/14/gate/gate-nova-python27/654a25b : SUCCESS in 18m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-full/2aa3148 : FAILURE in 52m 04s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-postgres-full/8e7d0d5 : SUCCESS in 49m 28s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-neutron-full/591ca0a : SUCCESS in 1h 06m 17s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-neutron-heat-slow/6a991d3 : SUCCESS in 24m 31s\n- gate-grenade-dsvm http://logs.openstack.org/65/121965/14/gate/gate-grenade-dsvm/20d86f1 : SUCCESS in 45m 35s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-large-ops/70b29b1 : SUCCESS in 24m 30s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-neutron-large-ops/3e52314 : SUCCESS in 24m 11s\n- gate-devstack-dsvm-cells http://logs.openstack.org/65/121965/14/gate/gate-devstack-dsvm-cells/7c00fe4 : SUCCESS in 16m 47s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/14/gate/gate-grenade-dsvm-partial-ncpu/ded968d : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":14},{"id":"7ece716785e9bebbbca154f05e03aa6b4c7c0929","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-10-29 12:56:13.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"67fcd1ed9225704b9842b75685d258b8b781ad27","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 14:14:16.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/14/check/gate-nova-pep8/fab87c6 : SUCCESS in 10m 43s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/14/check/gate-nova-docs/1cf7ae5/doc/build/html/ : SUCCESS in 8m 18s\n- gate-nova-python27 http://logs.openstack.org/65/121965/14/check/gate-nova-python27/f0fc4c5 : SUCCESS in 19m 26s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-full/58d10bf : SUCCESS in 57m 15s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-postgres-full/00b6858 : SUCCESS in 54m 38s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-neutron-full/a479ede : SUCCESS in 1h 11m 55s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-neutron-heat-slow/8d0871b : SUCCESS in 22m 24s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/14/check/check-grenade-dsvm/f3aa0bf : FAILURE in 48m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/14/check/gate-tempest-dsvm-large-ops/1c0bfd3 : SUCCESS in 25m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/14/check/gate-tempest-dsvm-neutron-large-ops/9189006 : SUCCESS in 26m 21s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/14/check/check-devstack-dsvm-cells/68dd7f8 : SUCCESS in 15m 56s\n- gate-nova-pylint http://logs.openstack.org/65/121965/14/check/gate-nova-pylint/09bd863 : SUCCESS in 20m 31s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/14/check/check-grenade-dsvm-partial-ncpu/6223825 : SUCCESS in 46m 23s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-ironic-pxe_ssh-nv/9deacb5 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":14},{"id":"267004587896f95d16784171a5c0894d38631409","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2014-10-29 14:17:08.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"227f2e2ca113022523a5eab8e6790509c6305c3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 15:53:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/65/121965/14/check/gate-nova-pep8/657e8da : SUCCESS in 9m 32s\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/14/check/gate-nova-docs/f17b77d/doc/build/html/ : SUCCESS in 10m 44s\n- gate-nova-python27 http://logs.openstack.org/65/121965/14/check/gate-nova-python27/773299b : SUCCESS in 19m 38s\n- check-tempest-dsvm-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-full/641e075 : SUCCESS in 1h 17m 56s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-postgres-full/b1dddaa : SUCCESS in 1h 14m 38s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-neutron-full/3717b84 : SUCCESS in 1h 23m 44s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-neutron-heat-slow/89cfd91 : SUCCESS in 40m 33s\n- check-grenade-dsvm http://logs.openstack.org/65/121965/14/check/check-grenade-dsvm/3bd72e7 : SUCCESS in 53m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/14/check/gate-tempest-dsvm-large-ops/10d0740 : SUCCESS in 27m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/14/check/gate-tempest-dsvm-neutron-large-ops/8d19a50 : SUCCESS in 27m 21s\n- check-devstack-dsvm-cells http://logs.openstack.org/65/121965/14/check/check-devstack-dsvm-cells/010c56e : SUCCESS in 28m 52s\n- gate-nova-pylint http://logs.openstack.org/65/121965/14/check/gate-nova-pylint/c55a572 : SUCCESS in 21m 14s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/14/check/check-grenade-dsvm-partial-ncpu/c4a3962 : SUCCESS in 54m 42s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/65/121965/14/check/check-tempest-dsvm-ironic-pxe_ssh-nv/93a432a : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":14},{"id":"56a896b593c7637ca09ec113333948eeeb204e96","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 15:53:25.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"1153cba2528c62651e7168dcfa0f5395efe19d99","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 18:02:23.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/65/121965/14/gate/gate-nova-docs/685f654/doc/build/html/ : SUCCESS in 9m 12s\n- gate-nova-pep8 http://logs.openstack.org/65/121965/14/gate/gate-nova-pep8/903ff42 : SUCCESS in 8m 12s\n- gate-nova-python27 http://logs.openstack.org/65/121965/14/gate/gate-nova-python27/ec7bd21 : SUCCESS in 14m 42s\n- gate-tempest-dsvm-full http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-full/da434c6 : SUCCESS in 45m 48s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-postgres-full/f751e8f : SUCCESS in 46m 06s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-neutron-full/e5b074e : SUCCESS in 59m 21s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-neutron-heat-slow/7e920c8 : SUCCESS in 26m 46s\n- gate-grenade-dsvm http://logs.openstack.org/65/121965/14/gate/gate-grenade-dsvm/231d0ce : SUCCESS in 44m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-large-ops/b22f071 : SUCCESS in 21m 10s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/65/121965/14/gate/gate-tempest-dsvm-neutron-large-ops/d52a3a3 : SUCCESS in 19m 01s\n- gate-devstack-dsvm-cells http://logs.openstack.org/65/121965/14/gate/gate-devstack-dsvm-cells/9c55688 : SUCCESS in 13m 19s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/65/121965/14/gate/gate-grenade-dsvm-partial-ncpu/75b882e : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":14},{"id":"891e174f21eb96bd32812cab91a2adbfe21f4000","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 18:02:26.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"36cfe6dce403a7dde6c5c12c8c6486f6048c9d35","revisions":{"8949228ba47add206c1b2f37529e6fa0d0267adb":{"kind":"REWORK","_number":1,"created":"2014-09-16 19:50:44.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/1"}}},"commit":{"parents":[{"commit":"1b45f4574c85288a6fd67a83ffd8c798d436875b","subject":"Split up libvirt volume\u0027s connect_volume method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b45f4574c85288a6fd67a83ffd8c798d436875b"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-16 19:18:55.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8949228ba47add206c1b2f37529e6fa0d0267adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8949228ba47add206c1b2f37529e6fa0d0267adb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1b45f4574c85288a6fd67a83ffd8c798d436875b","is_merged_in_target_branch":true,"change_id":"I5d300f8fe99f714d63d51197dc89f556c3a5d43b","change_number":116998,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fa261b1fabf9a4a3617ab49aaf36b976728f1545":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2014-09-17 16:45:56.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/2"}}},"commit":{"parents":[{"commit":"c52f0a8b2432de858f5592c298d74afc9984d1b4","subject":"Merge \"Split up libvirt volume\u0027s connect_volume method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c52f0a8b2432de858f5592c298d74afc9984d1b4"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-17 16:45:55.000000000","tz":0},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa261b1fabf9a4a3617ab49aaf36b976728f1545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa261b1fabf9a4a3617ab49aaf36b976728f1545"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c52f0a8b2432de858f5592c298d74afc9984d1b4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"731c0f57023f6bc527b86f9fbb48035753c7db15":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2014-09-18 01:20:31.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/3"}}},"commit":{"parents":[{"commit":"5becdc168a5526edc3f7f5679a13de4d6211721e","subject":"Merge \"Network: enable instance deletion when dhcp release fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5becdc168a5526edc3f7f5679a13de4d6211721e"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-18 01:20:05.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/731c0f57023f6bc527b86f9fbb48035753c7db15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/731c0f57023f6bc527b86f9fbb48035753c7db15"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5becdc168a5526edc3f7f5679a13de4d6211721e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"886739eecd236db744fb8109e9baa08eb6dd7676":{"kind":"REWORK","_number":4,"created":"2014-09-18 02:26:21.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/4"}}},"commit":{"parents":[{"commit":"5becdc168a5526edc3f7f5679a13de4d6211721e","subject":"Merge \"Network: enable instance deletion when dhcp release fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5becdc168a5526edc3f7f5679a13de4d6211721e"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-18 02:25:51.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/886739eecd236db744fb8109e9baa08eb6dd7676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/886739eecd236db744fb8109e9baa08eb6dd7676"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5becdc168a5526edc3f7f5679a13de4d6211721e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"767a146a48b8b4f8b03e81e47f172163189c75de":{"kind":"REWORK","_number":5,"created":"2014-09-18 02:55:54.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/5"}}},"commit":{"parents":[{"commit":"5becdc168a5526edc3f7f5679a13de4d6211721e","subject":"Merge \"Network: enable instance deletion when dhcp release fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5becdc168a5526edc3f7f5679a13de4d6211721e"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-18 02:55:36.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/767a146a48b8b4f8b03e81e47f172163189c75de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/767a146a48b8b4f8b03e81e47f172163189c75de"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5becdc168a5526edc3f7f5679a13de4d6211721e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ee8c6e94f29980c37efc6e9af520d8a1d9a6f8e3":{"kind":"REWORK","_number":6,"created":"2014-09-18 04:50:33.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/6"}}},"commit":{"parents":[{"commit":"5becdc168a5526edc3f7f5679a13de4d6211721e","subject":"Merge \"Network: enable instance deletion when dhcp release fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5becdc168a5526edc3f7f5679a13de4d6211721e"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-18 04:47:23.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nThe following patch also removes the call to _connect_volume\nin _create_domain_and_network, which is used to cache the\ndevice_path in connection_info, and adds the logic to cache\nthe device_path in _get_guest_storage_config. This removes\nan extra call connect_volume during spawn.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee8c6e94f29980c37efc6e9af520d8a1d9a6f8e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee8c6e94f29980c37efc6e9af520d8a1d9a6f8e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5becdc168a5526edc3f7f5679a13de4d6211721e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6faf754eb7b712424bbaecf84b8e924de2b050e8":{"kind":"REWORK","_number":7,"created":"2014-09-18 21:06:48.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/7"}}},"commit":{"parents":[{"commit":"f804ce594a9ba0617d838b496557111ef5b189cb","subject":"Merge \"Assume VNIC_NORMAL if binding:vnic_type not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f804ce594a9ba0617d838b496557111ef5b189cb"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-18 20:42:38.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nThe following patch also removes the call to _connect_volume\nin _create_domain_and_network, which is used to cache the\ndevice_path in connection_info, and adds the logic to cache\nthe device_path in _get_guest_storage_config. This removes\nan extra call connect_volume during spawn.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6faf754eb7b712424bbaecf84b8e924de2b050e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6faf754eb7b712424bbaecf84b8e924de2b050e8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f804ce594a9ba0617d838b496557111ef5b189cb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e7705bc5024992d427341e6bcac33e6a1c6caed8":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2014-09-18 21:11:07.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/8"}}},"commit":{"parents":[{"commit":"f804ce594a9ba0617d838b496557111ef5b189cb","subject":"Merge \"Assume VNIC_NORMAL if binding:vnic_type not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f804ce594a9ba0617d838b496557111ef5b189cb"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-18 21:07:53.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method is split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nThe following patch also removes the call to _connect_volume\nin _create_domain_and_network, which is used to cache the\ndevice_path in connection_info, and adds the logic to cache\nthe device_path in _get_volume_config. This will make sure\nthat the device_path in connection_info is set whenever\n_get_volume_config is called.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7705bc5024992d427341e6bcac33e6a1c6caed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7705bc5024992d427341e6bcac33e6a1c6caed8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f804ce594a9ba0617d838b496557111ef5b189cb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3256f520fe16a90c18ca83754241383ff3092dc9":{"kind":"REWORK","_number":9,"created":"2014-09-22 04:54:57.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/9"}}},"commit":{"parents":[{"commit":"e5fbc99af812ff6d55af92c12e7213c50b11b730","subject":"Merge \"libvirt: convert encrypted LVM test to avoid DB usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e5fbc99af812ff6d55af92c12e7213c50b11b730"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-22 04:41:48.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method were split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nThe following patch also updates\nconnection_info[\u0027data\u0027][\u0027device_path\u0027] on calls to libvirt\nvolume\u0027s connect_volume, so that it is properly saved to the\ndatabase on volume attach and swap.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3256f520fe16a90c18ca83754241383ff3092dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3256f520fe16a90c18ca83754241383ff3092dc9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e5fbc99af812ff6d55af92c12e7213c50b11b730","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6159cf0f7faddb43b39d9886fd78994070c9b7b6":{"kind":"REWORK","_number":10,"created":"2014-09-30 01:37:28.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/10"}}},"commit":{"parents":[{"commit":"e7469a8c1bb9e8bd1c8fb593170b6ada7a241112","subject":"Merge \"libvirt: log exception info when interface detach failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7469a8c1bb9e8bd1c8fb593170b6ada7a241112"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-30 01:35:31.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method were split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nThe following patch also updates\nconnection_info[\u0027data\u0027][\u0027device_path\u0027] on calls to libvirt\nvolume\u0027s connect_volume, so that it is properly saved to the\ndatabase on volume attach and swap.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6159cf0f7faddb43b39d9886fd78994070c9b7b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6159cf0f7faddb43b39d9886fd78994070c9b7b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e7469a8c1bb9e8bd1c8fb593170b6ada7a241112","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"66345b16443ceaa9c1df3f88fd6fb091f9c32c3a":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2014-09-30 01:39:20.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/11"}}},"commit":{"parents":[{"commit":"e7469a8c1bb9e8bd1c8fb593170b6ada7a241112","subject":"Merge \"libvirt: log exception info when interface detach failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7469a8c1bb9e8bd1c8fb593170b6ada7a241112"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-30 01:38:28.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method were split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66345b16443ceaa9c1df3f88fd6fb091f9c32c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66345b16443ceaa9c1df3f88fd6fb091f9c32c3a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e7469a8c1bb9e8bd1c8fb593170b6ada7a241112","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1570d818a67cef9a5f78c3c208885befb1f9640f":{"kind":"REWORK","_number":12,"created":"2014-09-30 15:25:00.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/12"}}},"commit":{"parents":[{"commit":"f5b37bac2eec89003cef6220722cc527a00ae7ee","subject":"Merge \"libvirt: Make sure NUMA cell memory is in Kb in XML\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5b37bac2eec89003cef6220722cc527a00ae7ee"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-30 15:24:39.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method were split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1570d818a67cef9a5f78c3c208885befb1f9640f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1570d818a67cef9a5f78c3c208885befb1f9640f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5b37bac2eec89003cef6220722cc527a00ae7ee","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1fbd7a170975aee52fc4be0b26d7f334532f3f9c":{"kind":"REWORK","_number":13,"created":"2014-09-30 19:04:44.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/13"}}},"commit":{"parents":[{"commit":"f5b37bac2eec89003cef6220722cc527a00ae7ee","subject":"Merge \"libvirt: Make sure NUMA cell memory is in Kb in XML\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5b37bac2eec89003cef6220722cc527a00ae7ee"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-30 18:37:43.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method were split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fbd7a170975aee52fc4be0b26d7f334532f3f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fbd7a170975aee52fc4be0b26d7f334532f3f9c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5b37bac2eec89003cef6220722cc527a00ae7ee","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"36cfe6dce403a7dde6c5c12c8c6486f6048c9d35":{"kind":"REWORK","_number":14,"created":"2014-10-07 18:50:07.000000000","uploader":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"ref":"refs/changes/65/121965/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/121965/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/121965/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/121965/14"}}},"commit":{"parents":[{"commit":"55bd7545042f40430e10d17033c7bc84badc1210","subject":"Merge \"libvirt: convert conn test case to avoid DB usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55bd7545042f40430e10d17033c7bc84badc1210"}]}],"author":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-09-15 18:54:40.000000000","tz":-240},"committer":{"name":"Thang Pham","email":"thang.g.pham@gmail.com","date":"2014-10-07 18:17:11.000000000","tz":-240},"subject":"Make separate calls to libvirt volume","message":"Make separate calls to libvirt volume\n\nThis is a follow up patch to\nhttps://review.openstack.org/#/c/116998/, where libvirt\nvolume\u0027s connect_volume method were split into get_config and\nconnect_volume. The following patch changes libvirt driver\nto call the appropriate volume method, get_config or\nconnect_volume, instead of the old monolithic call to\nconnect_volume. connect_volume will no longer need to return\nthe libvirt xml after this switch over. This will improve\nperformance and eliminate possible race conditions.\n\nChange-Id: I0d0d7ba98c53ad7509c28bb4866c488c0a049d2f\nCloses-Bug: #1359617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/36cfe6dce403a7dde6c5c12c8c6486f6048c9d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/36cfe6dce403a7dde6c5c12c8c6486f6048c9d35"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"55bd7545042f40430e10d17033c7bc84badc1210","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
