)]}'
{"id":"openstack%2Fnova~125492","triplet_id":"openstack%2Fnova~stable%2Ficehouse~I7598afbf0dc3c527471af34224003d28e64daaff","project":"openstack/nova","branch":"stable/icehouse","topic":"bug/1359138","hashtags":[],"change_id":"I7598afbf0dc3c527471af34224003d28e64daaff","subject":"VMWare: Fix VM leak when deletion of VM during resizing","status":"MERGED","created":"2014-10-01 20:23:44.000000000","updated":"2014-10-14 10:52:12.000000000","submitted":"2014-10-14 10:52:09.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ffa0b08924569e51cd07ab4f8b8a6d864a32950f","_number":125492,"virtual_id_number":125492,"owner":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":2,"date":"2014-10-14 10:52:09.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"date":"2014-10-09 12:00:10.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2014-10-13 16:35:47.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"value":0,"date":"2014-10-09 12:28:24.000000000","_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-10-13 14:59:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-10-13 15:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2014-10-13 14:49:59.000000000","_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-10-13 15:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-10-01 20:23:46.000000000","updated_by":{"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},"reviewer":{"_account_id":7148,"name":"Zhu Zhu","email":"bjzzu.zz@gmail.com","username":"zhuzhubj"},"state":"REVIEWER"},{"updated":"2014-10-01 22:11:01.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-09 12:00:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-10-09 12:28:24.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2014-10-13 14:49:59.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2014-10-13 14:59:49.000000000","updated_by":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"reviewer":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"state":"REVIEWER"},{"updated":"2014-10-13 15:39:45.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2014-10-13 16:35:47.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-10-14 10:52:09.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4725587ae728a58b313e84370badce893ba3133b","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2014-10-01 20:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b101e022e479670f663cecbbc691df6e39366dd0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-01 21:51:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/92/125492/1/check/gate-nova-pep8/b2a8748 : SUCCESS in 9m 39s\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/1/check/gate-nova-docs/f4969dc/doc/build/html/ : SUCCESS in 9m 40s\n- gate-nova-python26 http://logs.openstack.org/92/125492/1/check/gate-nova-python26/f33c997 : FAILURE in 15m 08s\n- gate-nova-python27 http://logs.openstack.org/92/125492/1/check/gate-nova-python27/cec94ec : FAILURE in 16m 45s\n- check-tempest-dsvm-full http://logs.openstack.org/92/125492/1/check/check-tempest-dsvm-full/9b451d5 : FAILURE in 57m 13s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/1/check/check-tempest-dsvm-postgres-full/f569a0a : SUCCESS in 1h 12m 24s\n- check-tempest-dsvm-neutron http://logs.openstack.org/92/125492/1/check/check-tempest-dsvm-neutron/49fcce2 : SUCCESS in 1h 10m 27s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/1/check/check-tempest-dsvm-neutron-heat-slow/05a5e74 : FAILURE in 36m 24s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/1/check/gate-tempest-dsvm-large-ops/f64301f : SUCCESS in 36m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/1/check/gate-tempest-dsvm-neutron-large-ops/fe77f82 : SUCCESS in 19m 20s\n- check-devstack-dsvm-cells http://logs.openstack.org/92/125492/1/check/check-devstack-dsvm-cells/4588f6e : SUCCESS in 15m 40s\n- gate-nova-pylint http://logs.openstack.org/92/125492/1/check/gate-nova-pylint/ef28753 : FAILURE in 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84dd94d2c71524b45e6291eec28f0d0a60388281","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-10-01 21:51:57.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: unrecognized error\n- check-tempest-dsvm-neutron-heat-slow: https://bugs.launchpad.net/bugs/1374175\n- gate-nova-python26: unrecognized error\n- gate-nova-python27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"bbfad65387637270c24b789b1312b8f3a6d7523d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-10-01 22:11:01.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/125492/1/ : Test run failed. in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4b9d976342c49000efb130fc53320d024e1d0210","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-10-02 10:54:53.000000000","message":"Patch Set 1: Code-Review-2\n\nThis fix breaks MS","accounts_in_message":[],"_revision_number":1},{"id":"5530d762ea26f0a3974c8142905cd376506f2000","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-10-02 13:04:26.000000000","message":"Patch Set 1:\n\n@Gary: you mean that breaks MS guests under the VMWare driver ? Since this is a backport, how do we avoid that unfortunate outcome in master branch ?","accounts_in_message":[],"_revision_number":1},{"id":"a2ff99e1750661c56762d13afc67c5d3bf81735d","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-10-06 14:27:46.000000000","message":"Patch Set 1:\n\nPlease see https://review.openstack.org/#/c/126177/\nWe need to also add this patch.\nI will base this above this one - unless you think that we should fold them in together","accounts_in_message":[],"_revision_number":1},{"id":"c5a96c992fce5405cef0a7368cc1536f2320088b","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-10-07 09:56:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f3ce535fb681ef677750a98bba685200b87fc63c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-07 10:58:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/92/125492/2/check/gate-nova-pep8/bb32316 : SUCCESS in 9m 52s\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/2/check/gate-nova-docs/ccc9608/doc/build/html/ : SUCCESS in 7m 59s\n- gate-nova-python26 http://logs.openstack.org/92/125492/2/check/gate-nova-python26/bef6d53 : SUCCESS in 15m 50s\n- gate-nova-python27 http://logs.openstack.org/92/125492/2/check/gate-nova-python27/eebe83b : SUCCESS in 15m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/92/125492/2/check/check-tempest-dsvm-full/a7cae3b : SUCCESS in 53m 26s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/2/check/check-tempest-dsvm-postgres-full/a27800f : SUCCESS in 47m 14s\n- check-tempest-dsvm-neutron http://logs.openstack.org/92/125492/2/check/check-tempest-dsvm-neutron/b04d9bf : SUCCESS in 55m 56s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/2/check/check-tempest-dsvm-neutron-heat-slow/fa4cde9 : SUCCESS in 22m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/2/check/gate-tempest-dsvm-large-ops/2e33d7d : SUCCESS in 26m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/2/check/gate-tempest-dsvm-neutron-large-ops/2a08c14 : SUCCESS in 23m 02s\n- check-devstack-dsvm-cells http://logs.openstack.org/92/125492/2/check/check-devstack-dsvm-cells/874eb3d : SUCCESS in 14m 40s\n- gate-nova-pylint http://logs.openstack.org/92/125492/2/check/gate-nova-pylint/55e3f9f : SUCCESS in 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"096f7713782d837f28917ebbcc57b832c00fd294","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2014-10-07 21:21:01.000000000","message":"Patch Set 2:\n\n@garyk Thanks! Is this change ready now ?","accounts_in_message":[],"_revision_number":2},{"id":"cc6771ee2c21514b59ea3f99edc1aaeeb09ad06c","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-10-08 09:52:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fc29ac9defdb3ab15d1b0e62d1fa03ad28a60e7c","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-10-08 09:53:48.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"5e928a2d2de0911cc9daa608298343a4bf36e1db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-08 11:04:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/92/125492/3/check/gate-nova-pep8/dff4350 : SUCCESS in 9m 45s\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/3/check/gate-nova-docs/6863eb9/doc/build/html/ : SUCCESS in 8m 54s\n- gate-nova-python26 http://logs.openstack.org/92/125492/3/check/gate-nova-python26/3137a03 : SUCCESS in 15m 55s\n- gate-nova-python27 http://logs.openstack.org/92/125492/3/check/gate-nova-python27/e775c39 : SUCCESS in 15m 17s\n- check-tempest-dsvm-full http://logs.openstack.org/92/125492/3/check/check-tempest-dsvm-full/c738ea1 : SUCCESS in 52m 29s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/3/check/check-tempest-dsvm-postgres-full/c6c0c9e : SUCCESS in 52m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/92/125492/3/check/check-tempest-dsvm-neutron/4910b95 : SUCCESS in 1h 02m 23s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/3/check/check-tempest-dsvm-neutron-heat-slow/ddeb496 : SUCCESS in 26m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/3/check/gate-tempest-dsvm-large-ops/7ca71d8 : SUCCESS in 22m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/3/check/gate-tempest-dsvm-neutron-large-ops/7395995 : SUCCESS in 22m 59s\n- check-devstack-dsvm-cells http://logs.openstack.org/92/125492/3/check/check-devstack-dsvm-cells/3b4cf0b : SUCCESS in 14m 11s\n- gate-nova-pylint http://logs.openstack.org/92/125492/3/check/gate-nova-pylint/cdaaacd : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a00f256548b8a6fb6be2c4400a7fddf19e7a1be","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-10-09 09:22:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nSee inline","accounts_in_message":[],"_revision_number":3},{"id":"66abdd1bd9382d430e4b5884f0e9b4464ef1f462","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2014-10-09 11:53:20.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nWill update this change soon","accounts_in_message":[],"_revision_number":3},{"id":"46e23c25cffdd6eb193b91023b07fba1661fcc86","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2014-10-09 11:59:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"29c7e769029f6c21db81c8c25951f352d684977d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-10-09 12:00:10.000000000","message":"Patch Set 4:\n\nPassed using PowerKVM driver. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For debugging, check https://wiki.openstack.org/wiki/PowerKVM#Debugging_PowerKVM_Failures","accounts_in_message":[],"_revision_number":4},{"id":"c895358606890dfe0259e66f707739c3843aee7a","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2014-10-09 12:28:24.000000000","message":"Patch Set 4:\n\nrecheck-pkvm","accounts_in_message":[],"_revision_number":4},{"id":"ee4b5b7f6392d28eb2b6da426bd8aabab1004006","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-10-09 12:28:33.000000000","message":"Patch Set 4:\n\nPassed using PowerKVM driver. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For debugging, check https://wiki.openstack.org/wiki/PowerKVM#Debugging_PowerKVM_Failures","accounts_in_message":[],"_revision_number":4},{"id":"e557308ee3faf5d40514c89b8914de85445d4127","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-09 13:01:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/92/125492/4/check/gate-nova-pep8/a25f837 : SUCCESS in 9m 26s\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/4/check/gate-nova-docs/d064361/doc/build/html/ : SUCCESS in 8m 34s\n- gate-nova-python26 http://logs.openstack.org/92/125492/4/check/gate-nova-python26/f555e3a : SUCCESS in 15m 24s\n- gate-nova-python27 http://logs.openstack.org/92/125492/4/check/gate-nova-python27/ffb032a : SUCCESS in 19m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-full/be80b78 : SUCCESS in 53m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-postgres-full/67441a7 : SUCCESS in 51m 41s\n- check-tempest-dsvm-neutron http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-neutron/51decfa : SUCCESS in 55m 53s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-neutron-heat-slow/1eb402d : SUCCESS in 22m 53s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/4/check/gate-tempest-dsvm-large-ops/66cabe3 : SUCCESS in 20m 57s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/4/check/gate-tempest-dsvm-neutron-large-ops/72f4241 : SUCCESS in 20m 27s\n- check-devstack-dsvm-cells http://logs.openstack.org/92/125492/4/check/check-devstack-dsvm-cells/c9ca25c : SUCCESS in 13m 23s\n- gate-nova-pylint http://logs.openstack.org/92/125492/4/check/gate-nova-pylint/9f8a1a8 : SUCCESS in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d66eef1d4426596e789976968bc59eb842bd023","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2014-10-09 13:25:12.000000000","message":"Patch Set 4:\n\nrecheck-pkvm","accounts_in_message":[],"_revision_number":4},{"id":"1d6cc888079abe1f59b78fcdd4622c269f74b6da","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-10-09 13:25:29.000000000","message":"Patch Set 4:\n\nPassed using PowerKVM driver. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For debugging, check https://wiki.openstack.org/wiki/PowerKVM#Debugging_PowerKVM_Failures\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/125492/4/check-ibm-tempest-dsvm-full/8e498e9 : NOT_REGISTERED (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/125492/4/check-ibm-tempest-dsvm-postgres-full/11ed11e : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d33d2c692724c01b049c3b9aa29c1fb52d4c483a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-09 14:25:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/92/125492/4/check/gate-nova-pep8/8f545ef : SUCCESS in 9m 25s\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/4/check/gate-nova-docs/4f408f8/doc/build/html/ : SUCCESS in 9m 44s\n- gate-nova-python26 http://logs.openstack.org/92/125492/4/check/gate-nova-python26/d43ba24 : SUCCESS in 16m 05s\n- gate-nova-python27 http://logs.openstack.org/92/125492/4/check/gate-nova-python27/407f9d1 : SUCCESS in 15m 10s\n- check-tempest-dsvm-full http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-full/f1941f9 : SUCCESS in 53m 02s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-postgres-full/79b6e33 : SUCCESS in 52m 38s\n- check-tempest-dsvm-neutron http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-neutron/98cdc20 : SUCCESS in 55m 53s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-neutron-heat-slow/8651094 : SUCCESS in 26m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/4/check/gate-tempest-dsvm-large-ops/cfdfcaf : SUCCESS in 23m 17s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/4/check/gate-tempest-dsvm-neutron-large-ops/9db7f72 : SUCCESS in 19m 07s\n- check-devstack-dsvm-cells http://logs.openstack.org/92/125492/4/check/check-devstack-dsvm-cells/6fc7614 : SUCCESS in 18m 17s\n- gate-nova-pylint http://logs.openstack.org/92/125492/4/check/gate-nova-pylint/9b904aa : SUCCESS in 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52cce0ca632ac0d0580e1b8fd1f5dbd44fdd66b2","author":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"date":"2014-10-09 15:19:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7e8e6a5bf4c9fe69bad19e3a613367c81455cc1c","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-10-13 14:49:59.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5e512289ff8138e2a4af70a31ac7cf3cd8d0c329","author":{"_account_id":7746,"name":"Dan Genin","email":"daniel.genin@jhuapl.edu","username":"DiG"},"date":"2014-10-13 14:59:49.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12a855ff43f41b8e62a35f25c1edd92b28f6abc1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-10-13 15:39:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"457377860fd41132d7646e6c89b0986e40892528","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-13 15:39:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"381064c5f1b91e7239ef96529594572154e44274","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-13 16:35:31.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/4/gate/gate-nova-docs/1a5e01b/doc/build/html/ : SUCCESS in 7m 32s\n- gate-nova-pep8 http://logs.openstack.org/92/125492/4/gate/gate-nova-pep8/4cae8c8 : SUCCESS in 9m 01s\n- gate-nova-python26 http://logs.openstack.org/92/125492/4/gate/gate-nova-python26/5c8a162 : SUCCESS in 15m 37s\n- gate-nova-python27 http://logs.openstack.org/92/125492/4/gate/gate-nova-python27/d7ccb04 : SUCCESS in 13m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-full/928b374 : SUCCESS in 49m 53s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-postgres-full/2c60044 : SUCCESS in 50m 03s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-neutron/dbb3f71 : FAILURE in 12m 53s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-neutron-heat-slow/4cfb14a : FAILURE in 12m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-large-ops/328bf08 : SUCCESS in 19m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-neutron-large-ops/7e264cf : FAILURE in 10m 40s\n- gate-devstack-dsvm-cells http://logs.openstack.org/92/125492/4/gate/gate-devstack-dsvm-cells/79efbff : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":4},{"id":"0b2d3667f764d3ad247d21ad9d1ce7dde45727e5","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-10-13 16:35:47.000000000","message":"Patch Set 4:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-heat-slow: https://bugs.launchpad.net/bugs/1379201\n- gate-tempest-dsvm-neutron-large-ops: https://bugs.launchpad.net/bugs/1379201\n- gate-tempest-dsvm-neutron: https://bugs.launchpad.net/bugs/1379201\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"ff5ba2d5f97405aaeeb6bd1cf9b29e7320539176","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-10-14 08:13:19.000000000","message":"Patch Set 4:\n\nrecheck bug 1379201","accounts_in_message":[],"_revision_number":4},{"id":"c697c2bd72e9466f217e256309d3d8f081ae9f3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-14 09:12:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/92/125492/4/check/gate-nova-pep8/8d664d7 : SUCCESS in 9m 39s\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/4/check/gate-nova-docs/7a11912/doc/build/html/ : SUCCESS in 8m 50s\n- gate-nova-python26 http://logs.openstack.org/92/125492/4/check/gate-nova-python26/91de9c2 : SUCCESS in 14m 54s\n- gate-nova-python27 http://logs.openstack.org/92/125492/4/check/gate-nova-python27/9809732 : SUCCESS in 13m 54s\n- check-tempest-dsvm-full http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-full/a53cb21 : SUCCESS in 52m 42s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-postgres-full/fd8fd9e : SUCCESS in 52m 44s\n- check-tempest-dsvm-neutron http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-neutron/94d7e3c : SUCCESS in 55m 08s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/4/check/check-tempest-dsvm-neutron-heat-slow/d4b0379 : SUCCESS in 20m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/4/check/gate-tempest-dsvm-large-ops/09820b7 : SUCCESS in 22m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/4/check/gate-tempest-dsvm-neutron-large-ops/6331602 : SUCCESS in 22m 04s\n- check-devstack-dsvm-cells http://logs.openstack.org/92/125492/4/check/check-devstack-dsvm-cells/7d3b8ff : SUCCESS in 15m 31s\n- gate-nova-pylint http://logs.openstack.org/92/125492/4/check/gate-nova-pylint/8d88c80 : SUCCESS in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"494b6300ba6a9a77978f55683cc52c79cc48d36c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-14 09:12:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"992dfcb4b4253120c48f08444e8c16c76d4a4fd8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-14 10:52:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded.\n\n- gate-nova-docs http://docs-draft.openstack.org/92/125492/4/gate/gate-nova-docs/20a2af2/doc/build/html/ : SUCCESS in 7m 37s\n- gate-nova-pep8 http://logs.openstack.org/92/125492/4/gate/gate-nova-pep8/4b18f21 : SUCCESS in 10m 28s\n- gate-nova-python26 http://logs.openstack.org/92/125492/4/gate/gate-nova-python26/8fe33e4 : SUCCESS in 15m 10s\n- gate-nova-python27 http://logs.openstack.org/92/125492/4/gate/gate-nova-python27/8761193 : SUCCESS in 13m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-full/801b771 : SUCCESS in 49m 25s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-postgres-full/aca8898 : SUCCESS in 46m 35s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-neutron/09578db : SUCCESS in 55m 25s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-neutron-heat-slow/9c9ab14 : SUCCESS in 20m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-large-ops/14516d7 : SUCCESS in 22m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/92/125492/4/gate/gate-tempest-dsvm-neutron-large-ops/f6aeb12 : SUCCESS in 20m 49s\n- gate-devstack-dsvm-cells http://logs.openstack.org/92/125492/4/gate/gate-devstack-dsvm-cells/a513c73 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1b8cda2a90385986d29d69e1055fcfa5e266f3a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-14 10:52:12.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d71445c7d2d2921d10a08f82330f0ab8ef4f7df2","revisions":{"8bc11709abaf2cdb3725d6e725ee53db5fd50f98":{"kind":"REWORK","_number":1,"created":"2014-10-01 20:23:44.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/92/125492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/92/125492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/92/125492/1"}}},"commit":{"parents":[{"commit":"c04b6ea0ef1770dc1b257d3018739cba1e7eadfb","subject":"Merge \"Adds tests for Hyper-V VM Utils\" into stable/icehouse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c04b6ea0ef1770dc1b257d3018739cba1e7eadfb"}]}],"author":{"name":"ZHU ZHU","email":"zhuzhubj@cn.ibm.com","date":"2014-09-03 08:59:13.000000000","tz":-300},"committer":{"name":"Tristan Cacqueray","email":"tristan.cacqueray@enovance.com","date":"2014-10-01 20:22:00.000000000","tz":0},"subject":"VMWare: Fix VM leak when deletion of VM during resizing","message":"VMWare: Fix VM leak when deletion of VM during resizing\n\nDuring the VM resizing, before VM arrive RESIZED state, driver\nmigrate_disk_and_power_off will initially rename orginal vm\n\u0027uuid\u0027 to be \u0027uuid-orig\u0027 and clone a new vm with \u0027uuid\u0027 name.\nWhen deletion VM is triggered at this time window, it wouldn\u0027t\nbe able to delete the VM uuid-orig in VCenter and so cause VM leak.\nAs VM task state will be set to \u0027deleting\u0027 and it can not be used to\ndetermine the resize migrating/migrated state, this fix will\nattempt to delete orig VM within destroy phase.\n\nConflicts:\n\tnova/tests/virt/vmwareapi/test_driver_api.py\n\tnova/virt/vmwareapi/vmops.py\n\nChange-Id: I7598afbf0dc3c527471af34224003d28e64daaff\nCloses-Bug: #1359138\n(cherry-picked from e1f8664c9fa83f77f5bb763ffcc3157905ed954c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bc11709abaf2cdb3725d6e725ee53db5fd50f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bc11709abaf2cdb3725d6e725ee53db5fd50f98"}]},"branch":"refs/heads/stable/icehouse"},"f9f2d4b1b5e107ac9b04f2fbc85de8ba5ff3cffd":{"kind":"REWORK","_number":2,"created":"2014-10-07 09:56:49.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/92/125492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/92/125492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/92/125492/2"}}},"commit":{"parents":[{"commit":"c04b6ea0ef1770dc1b257d3018739cba1e7eadfb","subject":"Merge \"Adds tests for Hyper-V VM Utils\" into stable/icehouse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c04b6ea0ef1770dc1b257d3018739cba1e7eadfb"}]}],"author":{"name":"ZHU ZHU","email":"zhuzhubj@cn.ibm.com","date":"2014-09-03 08:59:13.000000000","tz":-300},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2014-10-07 09:55:25.000000000","tz":-420},"subject":"VMWare: Fix VM leak when deletion of VM during resizing","message":"VMWare: Fix VM leak when deletion of VM during resizing\n\nDuring the VM resizing, before VM arrive RESIZED state, driver\nmigrate_disk_and_power_off will initially rename orginal vm\n\u0027uuid\u0027 to be \u0027uuid-orig\u0027 and clone a new vm with \u0027uuid\u0027 name.\nWhen deletion VM is triggered at this time window, it wouldn\u0027t\nbe able to delete the VM uuid-orig in VCenter and so cause VM leak.\nAs VM task state will be set to \u0027deleting\u0027 and it can not be used to\ndetermine the resize migrating/migrated state, this fix will\nattempt to delete orig VM within destroy phase.\n\nConflicts:\n\tnova/tests/virt/vmwareapi/test_driver_api.py\n\tnova/virt/vmwareapi/vmops.py\n\nCloses-Bug: #1359138\n\nNOTE: the aformentioned patch broke Minesweeper. The fix was also\ncherry picked from commit e464bc518e8590d59c2741948466777982ca3319.\nThis was to do two things:\n1. Solve the actual bug\n2. Ensure that the unit tests and Minesweeper passed\n\nChange-Id: I7598afbf0dc3c527471af34224003d28e64daaff\n(cherry-picked from e1f8664c9fa83f77f5bb763ffcc3157905ed954c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9f2d4b1b5e107ac9b04f2fbc85de8ba5ff3cffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9f2d4b1b5e107ac9b04f2fbc85de8ba5ff3cffd"}]},"branch":"refs/heads/stable/icehouse"},"14c07cb662b86b55fba86244cff2f94f15d47161":{"kind":"REWORK","_number":3,"created":"2014-10-08 09:52:40.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/92/125492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/92/125492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/92/125492/3"}}},"commit":{"parents":[{"commit":"c04b6ea0ef1770dc1b257d3018739cba1e7eadfb","subject":"Merge \"Adds tests for Hyper-V VM Utils\" into stable/icehouse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c04b6ea0ef1770dc1b257d3018739cba1e7eadfb"}]}],"author":{"name":"ZHU ZHU","email":"zhuzhubj@cn.ibm.com","date":"2014-09-03 08:59:13.000000000","tz":-300},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2014-10-08 09:51:59.000000000","tz":-420},"subject":"VMWare: Fix VM leak when deletion of VM during resizing","message":"VMWare: Fix VM leak when deletion of VM during resizing\n\nDuring the VM resizing, before VM arrive RESIZED state, driver\nmigrate_disk_and_power_off will initially rename orginal vm\n\u0027uuid\u0027 to be \u0027uuid-orig\u0027 and clone a new vm with \u0027uuid\u0027 name.\nWhen deletion VM is triggered at this time window, it wouldn\u0027t\nbe able to delete the VM uuid-orig in VCenter and so cause VM leak.\nAs VM task state will be set to \u0027deleting\u0027 and it can not be used to\ndetermine the resize migrating/migrated state, this fix will\nattempt to delete orig VM within destroy phase.\n\nConflicts:\n\tnova/tests/virt/vmwareapi/test_driver_api.py\n\tnova/virt/vmwareapi/vmops.py\n\nCloses-Bug: #1359138\n\nNOTE: the aformentioned patch broke Minesweeper. The fix was also\ncherry picked from commit e464bc518e8590d59c2741948466777982ca3319.\nThis was to do two things:\n1. Solve the actual bug\n2. Ensure that the unit tests and Minesweeper passed\n\nChange-Id: I7598afbf0dc3c527471af34224003d28e64daaff\n(cherry-picked from e1f8664c9fa83f77f5bb763ffcc3157905ed954c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14c07cb662b86b55fba86244cff2f94f15d47161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14c07cb662b86b55fba86244cff2f94f15d47161"}]},"branch":"refs/heads/stable/icehouse"},"d71445c7d2d2921d10a08f82330f0ab8ef4f7df2":{"kind":"REWORK","_number":4,"created":"2014-10-09 11:59:59.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/92/125492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/92/125492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/92/125492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/92/125492/4"}}},"commit":{"parents":[{"commit":"cb9cf67bbfd9826831b87eb7cfaeb9dbd5f984e1","subject":"Merge \"libvirt: Handle unsupported host capabilities\" into stable/icehouse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb9cf67bbfd9826831b87eb7cfaeb9dbd5f984e1"}]}],"author":{"name":"ZHU ZHU","email":"zhuzhubj@cn.ibm.com","date":"2014-09-03 08:59:13.000000000","tz":-300},"committer":{"name":"Tristan Cacqueray","email":"tristan.cacqueray@enovance.com","date":"2014-10-09 11:58:41.000000000","tz":0},"subject":"VMWare: Fix VM leak when deletion of VM during resizing","message":"VMWare: Fix VM leak when deletion of VM during resizing\n\nDuring the VM resizing, before VM arrive RESIZED state, driver\nmigrate_disk_and_power_off will initially rename orginal vm\n\u0027uuid\u0027 to be \u0027uuid-orig\u0027 and clone a new vm with \u0027uuid\u0027 name.\nWhen deletion VM is triggered at this time window, it wouldn\u0027t\nbe able to delete the VM uuid-orig in VCenter and so cause VM leak.\nAs VM task state will be set to \u0027deleting\u0027 and it can not be used to\ndetermine the resize migrating/migrated state, this fix will\nattempt to delete orig VM within destroy phase.\n\nConflicts:\n\tnova/tests/virt/vmwareapi/test_driver_api.py\n\tnova/virt/vmwareapi/vmops.py\n\nCloses-Bug: #1359138\n\nNOTE: the aformentioned patch broke Minesweeper. The fix was also\ncherry picked from commit e464bc518e8590d59c2741948466777982ca3319.\nThis was to do two things:\n1. Solve the actual bug\n2. Ensure that the unit tests and Minesweeper passed\n\nChange-Id: I7598afbf0dc3c527471af34224003d28e64daaff\n(cherry-picked from e1f8664c9fa83f77f5bb763ffcc3157905ed954c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d71445c7d2d2921d10a08f82330f0ab8ef4f7df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d71445c7d2d2921d10a08f82330f0ab8ef4f7df2"}]},"branch":"refs/heads/stable/icehouse"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
