)]}'
{"id":"openstack%2Fnova~133664","triplet_id":"openstack%2Fnova~master~I3e35598ec89b18d1121547caa37ec5dbc506cb06","project":"openstack/nova","branch":"master","topic":"bp/liberty-objects","hashtags":[],"change_id":"I3e35598ec89b18d1121547caa37ec5dbc506cb06","subject":"Remove and deprecate unused conductor service calls","status":"MERGED","created":"2014-11-11 12:10:42.000000000","updated":"2015-04-17 18:02:53.000000000","submitted":"2015-04-17 18:02:50.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7f3308fd2ec3fc21952ae15a01875f4dc8ff735","_number":133664,"virtual_id_number":133664,"owner":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2015-04-17 18:02:50.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-04-13 19:05:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"date":"2015-04-13 18:04:28.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2015-04-13 18:17:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-13 19:09:23.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"value":0,"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":2,"date":"2015-04-17 16:41:11.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":2,"date":"2015-04-14 16:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":1,"date":"2015-04-14 04:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-04-14 22:09:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"value":1,"date":"2015-04-15 16:38:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2015-04-13 18:03:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2015-04-17 16:41:11.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"value":0,"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"value":0,"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-11-11 13:45:07.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-01-27 01:49:47.000000000","updated_by":{"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},"reviewer":{"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},"state":"REVIEWER"},{"updated":"2015-01-27 13:17:28.000000000","updated_by":{"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},"reviewer":{"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},"state":"REVIEWER"},{"updated":"2015-02-17 21:28:45.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2015-02-27 22:57:54.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2015-03-23 18:15:23.000000000","updated_by":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-03-23 19:01:25.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-03-26 10:51:53.000000000","updated_by":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"reviewer":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"state":"REVIEWER"},{"updated":"2015-03-28 12:40:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-13 18:03:30.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2015-04-13 18:04:28.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-13 18:17:04.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-13 19:05:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-13 19:09:23.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-14 04:54:31.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"},{"updated":"2015-04-14 16:19:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2015-04-14 22:09:06.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2015-04-15 16:38:19.000000000","updated_by":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"reviewer":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"state":"REVIEWER"},{"updated":"2015-04-17 16:41:11.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2015-04-17 18:02:50.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c8af72365ed70eb08c7e7f9d2c4652eccac6acdb","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2014-11-11 12:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acf304802aecf00045aec88cac1f8abbf28244d5","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2014-11-11 12:34:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d7196262d2416412fde0c62da5c758f80a764c9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-11-11 12:55:42.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7d611aab05f34428b9b5d2e0c9ff5647/131007_devstack_export.log : SUCCESS in 16m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f88eb1f0ff0f4cea8160f586ff4ba0ca/user_001.log : SUCCESS in 19m 29s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e45bbe4fe639413197ef9da8bae1e47e/131007_devstack_export.log : SUCCESS in 16m 54s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/2/check/real-db-upgrade_nova_percona_user_001:th-percona/660c5a7c4f044df6ae564dfb475788d1/user_001.log : SUCCESS in 19m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/eb4e3059da8e407b9c20072d45770cb8/user_002.log : SUCCESS in 20m 20s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/2/check/real-db-upgrade_nova_percona_user_002:th-percona/0e4a22051e704faa93f1312878988965/user_002.log : SUCCESS in 17m 25s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"c48b8857e9d4ae58cbb4e955532b7004bed1e467","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-11 13:44:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/2/check/gate-nova-pep8/ddfb8ba : SUCCESS in 8m 15s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/2/check/gate-nova-docs/6d4a619/doc/build/html/ : SUCCESS in 8m 17s\n- gate-nova-python27 http://logs.openstack.org/64/133664/2/check/gate-nova-python27/c405129 : SUCCESS in 13m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/2/check/check-tempest-dsvm-full/b66a17c : FAILURE in 1h 02m 09s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/2/check/check-tempest-dsvm-postgres-full/6ccb52d : FAILURE in 1h 03m 37s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/2/check/check-tempest-dsvm-neutron-full/25ba1da : FAILURE in 1h 08m 09s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/64/133664/2/check/check-tempest-dsvm-neutron-heat-slow/7b71c0a : SUCCESS in 30m 43s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/2/check/check-grenade-dsvm/054fb9d : SUCCESS in 38m 38s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/2/check/gate-tempest-dsvm-large-ops/7f9665a : SUCCESS in 29m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/2/check/gate-tempest-dsvm-neutron-large-ops/439759e : SUCCESS in 28m 25s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/2/check/check-devstack-dsvm-cells/fafb091 : SUCCESS in 31m 45s\n- gate-nova-pylint http://logs.openstack.org/64/133664/2/check/gate-nova-pylint/ecf256c : SUCCESS in 14m 48s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/2/check/gate-nova-tox-functional/a95553a : FAILURE in 46s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/2/check/check-grenade-dsvm-partial-ncpu/672503c : SUCCESS in 1h 03m 15s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/64/133664/2/check/check-tempest-dsvm-ironic-pxe_ssh-nv/e62aa72 : FAILURE in 32m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d67859f6b74cc3f98007daa2fd5679d8d554bb64","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-11-11 13:45:07.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1311066\n- check-tempest-dsvm-ironic-pxe_ssh-nv: https://bugs.launchpad.net/bugs/1311066\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1311066\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1311066\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"f80405cfb0cf4990bca15e6b35e968454c51a4f5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-11-11 13:46:21.000000000","message":"Patch Set 2: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/2/30949/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"bc8d8eabfe16a93f36c3502300ea6968ca79288d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-11-11 13:51:24.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM.\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/2/check-ibm-tempest-dsvm-full/deb8618 : FAILURE in 1h 16m 33s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/2/check-ibm-tempest-dsvm-postgres-full/b9f6c16 : FAILURE in 54m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfe57c9b073530518d0f6b2482b2544319ed99f1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-11 13:52:20.000000000","message":"Patch Set 2:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/133664/2/ : Test run failed. in 1h 17m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3d498957d9b8087b859fe4d97284aeae9a030717","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2014-11-11 14:13:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d51f345ee8e84f628c7e7f22df401ad09f35c9eb","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-11-11 14:34:12.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d7ffd1e29b3349be85f71482ab1be43b/131007_devstack_export.log : SUCCESS in 16m 25s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/fb98355dda934fc6a06eeaf08f468a7b/user_001.log : SUCCESS in 17m 44s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/77a4b188ab294cf2a2b66ffe66c0af03/131007_devstack_export.log : SUCCESS in 16m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/3/check/real-db-upgrade_nova_percona_user_001:th-percona/a24ea182026a4e18aa8ce9f40bf7c18f/user_001.log : SUCCESS in 19m 56s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/eb89a9b606a44ad7add14854dd7915fc/user_002.log : SUCCESS in 17m 15s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/3/check/real-db-upgrade_nova_percona_user_002:th-percona/087b45c69bde4f93abdd5cb9972d022d/user_002.log : SUCCESS in 17m 41s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"90895c00a9192965366349ad9580a8fee61b343f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-11-11 15:13:42.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM.\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/3/check-ibm-tempest-dsvm-full/1dcd291 : SUCCESS in 58m 37s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/3/check-ibm-tempest-dsvm-postgres-full/9d76b7f : SUCCESS in 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a28408e4bd0b6b805a1ba06fc07227fb6b29ec20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-11 15:15:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/3/check/gate-nova-pep8/2694d27 : SUCCESS in 12m 21s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/3/check/gate-nova-docs/de0cde1/doc/build/html/ : SUCCESS in 8m 50s\n- gate-nova-python27 http://logs.openstack.org/64/133664/3/check/gate-nova-python27/2bfc97e : SUCCESS in 18m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/3/check/check-tempest-dsvm-full/8cb4dd6 : SUCCESS in 48m 53s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/3/check/check-tempest-dsvm-postgres-full/1fc8266 : SUCCESS in 50m 17s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/3/check/check-tempest-dsvm-neutron-full/87ece30 : SUCCESS in 56m 23s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/64/133664/3/check/check-tempest-dsvm-neutron-heat-slow/21aabf4 : SUCCESS in 19m 28s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/3/check/check-grenade-dsvm/52234ed : SUCCESS in 39m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/3/check/gate-tempest-dsvm-large-ops/f2abd6e : SUCCESS in 26m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/3/check/gate-tempest-dsvm-neutron-large-ops/0d02fcf : SUCCESS in 24m 19s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/3/check/check-devstack-dsvm-cells/3b03c0f : SUCCESS in 17m 39s\n- gate-nova-pylint http://logs.openstack.org/64/133664/3/check/gate-nova-pylint/2fe55d8 : SUCCESS in 22m 03s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/3/check/gate-nova-tox-functional/cebbbcd : FAILURE in 1m 02s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/3/check/check-grenade-dsvm-partial-ncpu/1a6eba4 : SUCCESS in 44m 06s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/64/133664/3/check/check-tempest-dsvm-ironic-pxe_ssh-nv/db9f70f : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":3},{"id":"99c0924fa0a07d723e7003047bb1a2a52912389b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-11-11 15:23:57.000000000","message":"Patch Set 3: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/3/30963/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"5867ba4a6d08e255e358ebcd5613e1c68f7257a9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-11 15:31:31.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/133664/3/ : Test run failed. in 1h 17m 47s","accounts_in_message":[],"_revision_number":3},{"id":"5955934765ed167240459a09d615146b0a3594e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-13 01:16:22.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"2f8d76094bc6c629b3c12ba82bb5f415f3ab3899","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-11-13 01:25:20.000000000","message":"Patch Set 3:\n\nBuild failed. Logs: http://208.91.1.172/logs/133664/3 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":3},{"id":"794f80d7e623dbb176a812c1bc75f7d5195ec443","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2014-11-13 08:15:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"38cc700f3b9a2542b15c2f53ae32d56e0e021b56","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-13 09:25:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/4/check/gate-nova-pep8/d5649d9 : SUCCESS in 11m 22s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/4/check/gate-nova-docs/d20b224/doc/build/html/ : SUCCESS in 6m 18s\n- gate-nova-python27 http://logs.openstack.org/64/133664/4/check/gate-nova-python27/159ab8f : SUCCESS in 13m 53s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/4/check/check-tempest-dsvm-full/6395e60 : SUCCESS in 43m 52s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/4/check/check-tempest-dsvm-postgres-full/965f87f : SUCCESS in 44m 05s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/4/check/check-tempest-dsvm-neutron-full/4330ed8 : SUCCESS in 1h 03m 56s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/64/133664/4/check/check-tempest-dsvm-neutron-heat-slow/62c87a2 : SUCCESS in 20m 31s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/4/check/check-grenade-dsvm/fde88ef : SUCCESS in 39m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/4/check/gate-tempest-dsvm-large-ops/93bab63 : SUCCESS in 23m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/4/check/gate-tempest-dsvm-neutron-large-ops/8dacdad : SUCCESS in 21m 50s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/4/check/check-devstack-dsvm-cells/e544152 : SUCCESS in 13m 38s\n- gate-nova-pylint http://logs.openstack.org/64/133664/4/check/gate-nova-pylint/61144ee : FAILURE in 14m 02s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/4/check/gate-nova-tox-functional/b6c4298 : FAILURE in 8m 39s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/4/check/check-grenade-dsvm-partial-ncpu/ce439c0 : SUCCESS in 47m 04s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/64/133664/4/check/check-tempest-dsvm-ironic-pxe_ssh-nv/fb271a1 : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":4},{"id":"97b3facb8cd8c8c59dc6f60c8578850a9242eafc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-11-13 10:50:14.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/254cc3818c194db6b1606c1d0049eccd/131007_devstack_export.log : SUCCESS in 18m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c0bbac3b1aa14e3b81566c626232bd08/user_001.log : SUCCESS in 19m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/19f8903047134124b39b1f2f2d1655e8/131007_devstack_export.log : SUCCESS in 17m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/4/check/real-db-upgrade_nova_percona_user_001:th-percona/74c3df5866c64a44ac4eb4435336db2d/user_001.log : SUCCESS in 18m 41s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/162d9db8d9034feeaaeba690c3e2bd06/user_002.log : SUCCESS in 20m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/4/check/real-db-upgrade_nova_percona_user_002:th-percona/1f2fd6b4499241d0aea88fde3fa082cb/user_002.log : SUCCESS in 19m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"5776e5869c16e0aca3f47c8928d3e7c3b042a8c4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-13 11:46:10.000000000","message":"Patch Set 4:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/133664/4/ : Test run failed. in 2h 00m 01s","accounts_in_message":[],"_revision_number":4},{"id":"f154b016f9f79aba85edd590161f90f50d3223cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-11-13 14:17:40.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM.\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/4/check-ibm-tempest-dsvm-full/7d97de7 : SUCCESS in 1h 02m 32s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/4/check-ibm-tempest-dsvm-postgres-full/07c98bc : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b6eae0294fb41cd0a80012d3b1f453b772dd93b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-11-13 14:45:41.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/4/31224/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"2f4a3ea24291ce93522d85c7c6ab89878d649633","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-11-14 12:12:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/4","accounts_in_message":[],"_revision_number":4},{"id":"511bc7accdd6e4b079d3d45c018862409db763c5","author":{"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},"date":"2014-11-14 16:10:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e038370a7ab633f387796c0a2236d259cedeb89e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-21 18:17:10.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"df0e908a14f30a78274abf8c30adeb0153b7e51a","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2014-11-22 10:10:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e1fc517b89015fa0ff70742377bdd08e7b71b414","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-11-22 10:33:46.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/100d163563ee4fcfa61666f58673b376/131007_devstack_export.log : SUCCESS in 18m 40s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8c36ffd70f8b498c84deca169b9ec8ae/user_001.log : SUCCESS in 22m 01s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5536f6373a1040119ab9066323415be2/131007_devstack_export.log : SUCCESS in 16m 28s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/5/check/real-db-upgrade_nova_percona_user_001:th-percona/b666b87f2b624bdd92ff3789b58dc952/user_001.log : SUCCESS in 18m 14s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/16a6a6878c84422481fd888ec9f01496/user_002.log : SUCCESS in 16m 58s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/5/check/real-db-upgrade_nova_percona_user_002:th-percona/e897e6fb33f44c5287d56a6ea3bbece8/user_002.log : SUCCESS in 17m 20s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"83c1939af43084f8443fbea768ecae0fefd4f05c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-11-22 10:54:56.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM.\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/5/check-ibm-tempest-dsvm-full/3a4f13f : SUCCESS in 43m 51s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/5/check-ibm-tempest-dsvm-postgres-full/8c05bf9 : SUCCESS in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e52d9ab6a1029860810c00be4159f469a907726f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-11-22 11:17:49.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/5/32715/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"fba1b6496715d55171bc4e7391ea445f3045b3bd","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-22 11:34:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/133664/5/ : Looks good to me in 1h 23m 35s","accounts_in_message":[],"_revision_number":5},{"id":"ea4a57bc83cd4a74f01db43bff376cb86dc9d0c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-22 16:43:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/5/check/gate-nova-pep8/729f50f : SUCCESS in 8m 04s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/5/check/gate-nova-docs/f68d6d3/doc/build/html/ : SUCCESS in 6m 31s\n- gate-nova-python27 http://logs.openstack.org/64/133664/5/check/gate-nova-python27/817cf90 : SUCCESS in 13m 17s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/5/check/check-tempest-dsvm-full/c376eb1 : SUCCESS in 40m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/5/check/check-tempest-dsvm-postgres-full/26e0098 : SUCCESS in 50m 45s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/5/check/check-tempest-dsvm-neutron-full/70b4615 : SUCCESS in 53m 17s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/64/133664/5/check/check-tempest-dsvm-neutron-heat-slow/28e59e6 : SUCCESS in 20m 30s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/5/check/check-grenade-dsvm/a5e7342 : SUCCESS in 36m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/5/check/gate-tempest-dsvm-large-ops/d18c27e : SUCCESS in 20m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/5/check/gate-tempest-dsvm-neutron-large-ops/252b037 : SUCCESS in 17m 32s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/5/check/check-devstack-dsvm-cells/5780d0a : SUCCESS in 12m 02s\n- gate-nova-pylint http://logs.openstack.org/64/133664/5/check/gate-nova-pylint/5fcbe2d : FAILURE in 18m 16s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/5/check/gate-nova-tox-functional/1311659 : FAILURE in 6m 21s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/5/check/check-grenade-dsvm-partial-ncpu/d06deb0 : SUCCESS in 34m 56s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/64/133664/5/check/check-tempest-dsvm-ironic-pxe_ssh-nv/71b6106 : SUCCESS in 25m 10s\n- check-tempest-dsvm-docker http://logs.openstack.org/64/133664/5/check/check-tempest-dsvm-docker/07e4500 : FAILURE in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6c0269b46d17d5ddb9e32f96be2645f9821d29c3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-11-24 19:38:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/5","accounts_in_message":[],"_revision_number":5},{"id":"e4c49c79d591137c509549acdb40e714cdce72e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-08 13:08:09.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"cc07c067b8d34ab339ee5f5efd6b5ece401a145c","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2014-12-08 21:17:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"59aa791fbe486178adcb87fcb19f62f26dd16882","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-12-08 22:03:50.000000000","message":"Patch Set 6: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/6/1671/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"5cb815ae45f7933b3b3bd0afaf6f88e592b9e973","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-08 22:15:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/6/check/gate-nova-pep8/7fff7b8 : SUCCESS in 8m 05s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/6/check/gate-nova-docs/d9f7838/doc/build/html/ : SUCCESS in 6m 10s\n- gate-nova-python27 http://logs.openstack.org/64/133664/6/check/gate-nova-python27/8fd5074 : SUCCESS in 11m 47s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/6/check/check-tempest-dsvm-full/196e4f4 : SUCCESS in 31m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/6/check/check-tempest-dsvm-postgres-full/7fa4cd7 : SUCCESS in 28m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/6/check/check-tempest-dsvm-neutron-full/3d29d0a : SUCCESS in 38m 27s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/6/check/check-grenade-dsvm/fee13b7 : SUCCESS in 24m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/6/check/gate-tempest-dsvm-large-ops/0646247 : SUCCESS in 19m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/6/check/gate-tempest-dsvm-neutron-large-ops/785d8bf : SUCCESS in 19m 30s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/6/check/check-devstack-dsvm-cells/f591645 : SUCCESS in 13m 26s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/6/check/gate-nova-tox-functional/87bc4a2 : SUCCESS in 16m 10s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/6/check/check-grenade-dsvm-partial-ncpu/ad7fd78 : SUCCESS in 28m 42s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/6/check/check-tempest-dsvm-ironic-pxe_ssh/ee2980e : SUCCESS in 25m 24s\n- check-tempest-dsvm-docker http://logs.openstack.org/64/133664/6/check/check-tempest-dsvm-docker/cc28403 : FAILURE in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3eed2fb8076d1764ac7dc6e7a3a69fdf583929e0","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-08 23:28:37.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3cad0c1969314e9d887eaef97702da7f/index.html : SUCCESS in 15m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/af4844fbf5dd436fbb07542ae93f6c30/index.html : SUCCESS in 17m 57s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4aff8f36203c442792c511b9c8657a7b/index.html : SUCCESS in 15m 38s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/6/check/real-db-upgrade_nova_percona_user_001:th-percona/cc7c3908a145480292b583e8c3a8d647/index.html : SUCCESS in 20m 26s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4ec35c2a88fe4323848208c219884eee/index.html : SUCCESS in 15m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/6/check/real-db-upgrade_nova_percona_user_002:th-percona/2fec913f80f44cd1a2c7a2f1a055ceec/index.html : SUCCESS in 16m 02s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"091b7bf4cfb07f7ccec388debf6f4cecda4ff2c7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-09 03:18:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/133664/6/ : Looks good to me in 1h 59m 48s","accounts_in_message":[],"_revision_number":6},{"id":"c097e21a0fb1f9a8c57f4d4c4b3d4070821f96e7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-12-12 03:49:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/6","accounts_in_message":[],"_revision_number":6},{"id":"a9cb4a57ad9e3c4b735a0470316288af2cce6da1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2014-12-13 19:42:40.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nSome questions inline. I assume you aim to remove only the things which got replaced by the service object-ification, but I see things like e.g. service_get_all removed which wasn\u0027t used in the first place by nova/service.py. So, I pointed out the remaining service-related conductor functions which seem also unused (double-check).","accounts_in_message":[],"_revision_number":6},{"id":"4e0f08f694f92d3a99e0592f5259ad4d671b8194","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-01-22 11:45:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1a4c40cb5bcc696b4a9d8ede821ccf4c1323ae84","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-01-22 11:54:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"00c079cbb5ae9ca64ca6869f8f32d614369bc1c2","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-01-22 12:00:17.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ebfc521a15d7a9d2cefe3fcd015f39308174362b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-01-22 12:42:03.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/8/check-ibm-tempest-dsvm-full/c40ebe4 : SUCCESS in 29m 14s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/8/check-ibm-tempest-dsvm-postgres-full/ba0441a : SUCCESS in 32m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2d390c2c3a1c458850658c3132a01d74c169526a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-01-22 12:56:29.000000000","message":"Patch Set 8: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/8/6107/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":8},{"id":"2e9431731a786469410c9d3c566141b8e91dad14","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-22 12:59:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/8/check/gate-nova-pep8/71fd38a/ : SUCCESS in 10m 11s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/8/check/gate-nova-docs/6f83c98//doc/build/html/ : SUCCESS in 11m 49s\n- gate-nova-python27 http://logs.openstack.org/64/133664/8/check/gate-nova-python27/705985b/ : SUCCESS in 15m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/8/check/check-tempest-dsvm-full/ef3f8a6/ : SUCCESS in 39m 10s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/8/check/check-tempest-dsvm-postgres-full/14f0c8f/ : SUCCESS in 48m 24s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/8/check/check-tempest-dsvm-neutron-full/81c4748/ : SUCCESS in 54m 58s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/8/check/check-grenade-dsvm/6f0f560/ : SUCCESS in 38m 32s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/8/check/gate-tempest-dsvm-large-ops/37f3763/ : SUCCESS in 22m 40s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/8/check/gate-tempest-dsvm-neutron-large-ops/0817b74/ : SUCCESS in 22m 48s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/8/check/check-devstack-dsvm-cells/70d587d/ : SUCCESS in 18m 06s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/8/check/gate-nova-tox-functional/7ef7e7e/ : SUCCESS in 15m 50s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/8/check/check-grenade-dsvm-partial-ncpu/ecccbdd/ : SUCCESS in 40m 19s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/8/check/check-tempest-dsvm-ironic-pxe_ssh/8bfc67c/ : SUCCESS in 28m 57s\n- check-tempest-dsvm-docker http://logs.openstack.org/64/133664/8/check/check-tempest-dsvm-docker/1dcc919/ : SUCCESS in 31m 07s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/8/check/check-tempest-dsvm-nova-v21-full/43d4cc7/ : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":8},{"id":"011f09890b4e2938fb524b94d0b976fbf3cd6305","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-01-22 14:57:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/133664/8/ : Looks good to me in 2h 11m 12s","accounts_in_message":[],"_revision_number":8},{"id":"948c2dfde97bd4d30d37080abe267282504afa01","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-01-22 15:00:58.000000000","message":"Patch Set 8:\n\nBuild Failed. Logs at http://198.175.100.33/133664/8","accounts_in_message":[],"_revision_number":8},{"id":"100331d1d6b8de18dbef16d2093ad818cf07ab32","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2015-01-22 21:01:45.000000000","message":"Patch Set 8: Code-Review+1\n\nCool, thanks Hans","accounts_in_message":[],"_revision_number":8},{"id":"7c950e35c771d15619bddf68659fd526db0007c8","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-01-27 01:44:25.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/ad567857c5d54cb7be6f5d5320fc28ae/index.html : SUCCESS in 17m 05s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8751d4d55a0c46a58b5ce6e3b7c2318a/index.html : SUCCESS in 21m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/705570a1098648d0bfef1793051479d4/index.html : SUCCESS in 30m 40s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/8/check/real-db-upgrade_nova_percona_user_001:th-percona/7d49c0c3058c4e1fb51125b9ce691157/index.html : SUCCESS in 24m 52s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a9ee3706258a431092c1a99d122a8e74/index.html : SUCCESS in 18m 49s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/8/check/real-db-upgrade_nova_percona_user_002:th-percona/2f7c54fd1f1b4b61b2caf0730f4012e4/index.html : SUCCESS in 18m 24s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"fcf58cd5e4beb4f6f3dbc6e7e163a6594b9a8239","author":{"_account_id":12952,"name":"ice yao","email":"yao3690093@gmail.com","username":"xiabing.yao"},"date":"2015-01-27 01:49:47.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"36dfce573dd60917929b251fa981f31256e772ff","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2015-01-27 02:15:37.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"9a866501ae13afdfb5b99b1bc19e4f07a3b21eb3","author":{"_account_id":9550,"name":"Eugeniya Kudryashova","email":"ekudryashova@mirantis.com","username":"ekudryashova"},"date":"2015-01-27 13:17:28.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"c5ff7f7fd3a6efea74ae1f943969a3fbe8fff43e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 14:48:43.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"567e70b7cb5d1c996cdc7e0a3a6ad465d3ae20a2","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-02-06 17:53:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6524d3ee6207d8523d730e41cbd213a8b9e82a8b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-06 18:15:30.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/c7fb03d99c2045f58fc0058f220ed288/index.html : SUCCESS in 20m 08s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4a010002601f4da494ff38289af47296/index.html : SUCCESS in 19m 45s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2fab1c43bd33459c81722f4e752f531d/index.html : SUCCESS in 16m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/9/check/real-db-upgrade_nova_percona_user_001:th-percona/62e95859481a4b5e8dc01dee789000da/index.html : SUCCESS in 20m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/09894eb21d544524b2f603ffb040fe87/index.html : SUCCESS in 20m 09s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/9/check/real-db-upgrade_nova_percona_user_002:th-percona/a12c2378d8fb44b483f6203e8d3fb631/index.html : SUCCESS in 19m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"b3a1c149d1942c9e2cb2dfa9cece6fefe217d5d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-02-06 18:41:55.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/9/check-ibm-tempest-dsvm-full/2cc47d9 : SUCCESS in 47m 29s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/9/check-ibm-tempest-dsvm-postgres-full/beb5c1d : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"63a7752383af3ce467a22f810da78261ab3f81ad","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-06 18:51:03.000000000","message":"Patch Set 9: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/9/8244/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":9},{"id":"e481201cc198887ef69cece48a3c22b996af2f49","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 19:10:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/9/check/gate-nova-pep8/5666603/ : SUCCESS in 10m 15s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/9/check/gate-nova-docs/45130f7//doc/build/html/ : SUCCESS in 10m 22s\n- gate-nova-python27 http://logs.openstack.org/64/133664/9/check/gate-nova-python27/7bc13b8/ : SUCCESS in 15m 06s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-full/0b64856/ : SUCCESS in 48m 44s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-postgres-full/5aff6a2/ : SUCCESS in 48m 39s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-neutron-full/03e3672/ : SUCCESS in 1h 04m 18s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/9/check/check-grenade-dsvm/d57cfcd/ : SUCCESS in 38m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/9/check/gate-tempest-dsvm-large-ops/ca009a1/ : SUCCESS in 23m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/9/check/gate-tempest-dsvm-neutron-large-ops/6278bd1/ : SUCCESS in 22m 14s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/9/check/check-devstack-dsvm-cells/509b0ee/ : SUCCESS in 18m 13s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/9/check/gate-nova-tox-functional/5c1670e/ : SUCCESS in 12m 00s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/9/check/check-grenade-dsvm-partial-ncpu/d416637/ : SUCCESS in 37m 00s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-ironic-pxe_ssh/0198540/ : SUCCESS in 34m 12s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-nova-v21-full/2e880e1/ : SUCCESS in 49m 04s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-cells/ff7391a/ : FAILURE in 34m 07s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/9/check/check-tempest-dsvm-full-ceph/570a697/ : SUCCESS in 44m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7847cc13c32409abc5caac9c38dd6fb39ac98d12","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-02-06 19:15:18.000000000","message":"Patch Set 9:\n\nBuild Successful. Logs at http://198.175.100.33/133664/9","accounts_in_message":[],"_revision_number":9},{"id":"f6379a4199a4111bb0d5d662a57b689bc58d7f0c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-02-06 19:33:36.000000000","message":"Patch Set 9:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/133664/9/ : Test run failed. in 1h 38m 46s","accounts_in_message":[],"_revision_number":9},{"id":"06cc5a140f397492f535faf03553af39499d9620","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2015-02-11 06:47:04.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"5075783445e69bacb96c3c4ab8e6f812c2529f3c","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-02-17 21:28:45.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7b07cff7b6c8dae5597c89d4f20b74f7f38153ba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-18 02:23:07.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"306e78356c5a0ff2a630be7098b0066374018635","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-02-18 09:18:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"663d3984bbe1278e6191133aef7e4fc009293299","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-02-18 09:28:05.000000000","message":"Patch Set 10:\n\nBuild Successful. Logs at http://198.175.100.33/133664/10","accounts_in_message":[],"_revision_number":10},{"id":"5319e78c10481a2ef35de57de092b35776f9c4db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-18 09:28:10.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"93616d194c0ec86f967ccdfdcac8b14728ee5498","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-18 09:28:15.000000000","message":"Patch Set 10: Verified+1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"9f79bd4868ecca071e2d213601faba6be88d6527","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-18 09:54:03.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/45c967cd2e5d495dbe1121afe3ac54e0/index.html : SUCCESS in 19m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/47f737a84c1e425288fe57c1045439f2/index.html : SUCCESS in 20m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1f66185ff4634083a468eef450dfc9a2/index.html : SUCCESS in 21m 09s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_percona_user_001:th-percona/2b0a2bc8e266486c895140e005ed3fa3/index.html : SUCCESS in 24m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/90669f23b4274aa790d0f68e36fa9b33/index.html : SUCCESS in 20m 22s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_percona_user_002:th-percona/d6dbb2d04cd44da0a121fb1b5f1e876e/index.html : SUCCESS in 21m 19s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"612b55b1b26ff0fd7aa9eb6f894277d2e8c956af","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-02-18 09:54:41.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/10/check-ibm-tempest-dsvm-full/e2d9dfc : SUCCESS in 35m 24s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/10/check-ibm-tempest-dsvm-postgres-full/380298b : SUCCESS in 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bb380a90e30890a0ca583d8af76f62d52c8bd191","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-18 10:18:41.000000000","message":"Patch Set 10: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/10/9719/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":10},{"id":"f233386106fa8b282ae2b64d07c64d9e9d1dd351","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-02-18 10:32:34.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a9f0c61d7ba0074adc1f4cdc67437468385eb64e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-02-18 11:25:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/10","accounts_in_message":[],"_revision_number":10},{"id":"fbc55daa8ad4263b578b2cf3ecf9958b23beac37","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-18 11:53:17.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/10/check/gate-nova-pep8/8177878/ : SUCCESS in 10m 29s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/10/check/gate-nova-docs/858eef5//doc/build/html/ : SUCCESS in 11m 37s\n- gate-nova-python27 http://logs.openstack.org/64/133664/10/check/gate-nova-python27/2fd96e1/ : SUCCESS in 15m 11s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-full/798ce49/ : FAILURE in 6m 59s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-postgres-full/5d6f0bd/ : SUCCESS in 54m 42s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-neutron-full/53ab9e4/ : SUCCESS in 1h 05m 04s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/10/check/check-grenade-dsvm/8379c41/ : SUCCESS in 44m 27s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/10/check/gate-tempest-dsvm-large-ops/96b790e/ : SUCCESS in 27m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/10/check/gate-tempest-dsvm-neutron-large-ops/972384d/ : SUCCESS in 24m 40s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/10/check/check-devstack-dsvm-cells/1b777c3/ : SUCCESS in 19m 47s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/10/check/gate-nova-tox-functional/bf7c2ec/ : SUCCESS in 14m 51s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/10/check/check-grenade-dsvm-partial-ncpu/ff6d6af/ : SUCCESS in 45m 33s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-ironic-pxe_ssh/68ab0c0/ : SUCCESS in 32m 17s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-nova-v21-full/c786126/ : SUCCESS in 49m 20s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-cells/d8b99d5/ : FAILURE in 33m 53s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-full-ceph/d75d369/ : SUCCESS in 46m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"60cf5c88530e62b3231930c56e75e968765e1568","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-02-18 12:11:45.000000000","message":"Patch Set 10:\n\nrecheck bug 1301820","accounts_in_message":[],"_revision_number":10},{"id":"db8c975cc0583039a2cb102b794e6909cb7c04d9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-18 12:46:07.000000000","message":"Patch Set 10:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9139da89be944c1fad7130348f1400d8/index.html : SUCCESS in 20m 10s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ff5de1e38a294f9c94317a4eef018ddc/index.html : SUCCESS in 18m 50s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5a5af170edc84d1db726fc402ac86af9/index.html : SUCCESS in 20m 42s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_percona_user_001:th-percona/c086155fa1e24fcdb9896612eaa7c6c1/index.html : SUCCESS in 24m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ff4aa6fe14ca4ab2a82c7469804c8198/index.html : SUCCESS in 20m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/10/check/real-db-upgrade_nova_percona_user_002:th-percona/c414f389cfd246acb5ba74c303dade2f/index.html : SUCCESS in 17m 00s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"ee7a9cf081023c39cd202635402a3ae0d4d92ad5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-18 13:36:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/10/check/gate-nova-pep8/103cf87/ : SUCCESS in 11m 25s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/10/check/gate-nova-docs/b5b64bf//doc/build/html/ : SUCCESS in 11m 28s\n- gate-nova-python27 http://logs.openstack.org/64/133664/10/check/gate-nova-python27/2ff0e7b/ : SUCCESS in 12m 35s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-full/c3baceb/ : SUCCESS in 44m 16s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-postgres-full/f6d8663/ : SUCCESS in 36m 31s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-neutron-full/a3fe288/ : SUCCESS in 1h 14m 24s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/10/check/check-grenade-dsvm/d8730ef/ : SUCCESS in 37m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/10/check/gate-tempest-dsvm-large-ops/f012cee/ : SUCCESS in 20m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/10/check/gate-tempest-dsvm-neutron-large-ops/6b92d34/ : SUCCESS in 20m 01s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/10/check/check-devstack-dsvm-cells/4491c27/ : SUCCESS in 14m 47s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/10/check/gate-nova-tox-functional/f007563/ : SUCCESS in 15m 29s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/10/check/check-grenade-dsvm-partial-ncpu/6c3b18b/ : SUCCESS in 36m 40s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-ironic-pxe_ssh/a9e8707/ : SUCCESS in 47m 54s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-nova-v21-full/d3d5017/ : SUCCESS in 45m 11s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-cells/968ca63/ : FAILURE in 35m 24s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/10/check/check-tempest-dsvm-full-ceph/35519d0/ : SUCCESS in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ce075dd4a5c5d98d1cef4a6ea81ca18aa670f90a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-02-18 14:39:09.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c3835999cc22396a8f8b9f48b8bf24b151539c79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-23 01:53:59.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"79025f861c93b65dc045d757ea6e28fef6b6a5ae","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-02-23 08:08:25.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c6f19eedc21e385b68d32df4ade38406917b8773","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-23 08:43:12.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1f7ca0d1218146c8a3e01e5c52414a23/index.html : SUCCESS in 20m 29s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ee882b8a2f5646ee9da4811279121c89/index.html : SUCCESS in 22m 32s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ddf06c319f3b441a8135a9f196c96c97/index.html : SUCCESS in 20m 24s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/11/check/real-db-upgrade_nova_percona_user_001:th-percona/0c8e450f85a74917b240136d39a0f03c/index.html : SUCCESS in 23m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/948d5b0d51344199a5c8d418fd0402dd/index.html : SUCCESS in 20m 29s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/11/check/real-db-upgrade_nova_percona_user_002:th-percona/6629039f7b9445b8b45bec44ce4a3a21/index.html : SUCCESS in 20m 29s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"32215f99a30310836590fd8c71a2ab9e8c48ffe0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-23 08:43:20.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"1f88244a50ac2cb345f53d56622c1f57b9281413","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-23 08:43:23.000000000","message":"Patch Set 11: Verified+1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"bf53a32b03aba2542f1f5796335ef56d0311b5b9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-23 09:12:31.000000000","message":"Patch Set 11: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/11/10204/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":11},{"id":"c4aa5fa0cd14e0aeb250055a39f09219216907f4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-02-23 14:21:30.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/11/check-ibm-tempest-dsvm-full/5965f7a : SUCCESS in 37m 49s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/11/check-ibm-tempest-dsvm-postgres-full/63b4192 : SUCCESS in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"237463571130408e8cdcb3b8212332c84690f853","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-02-23 22:05:21.000000000","message":"Patch Set 11:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/133664/11/ : Test run failed. in 1h 05m 47s","accounts_in_message":[],"_revision_number":11},{"id":"e84c65a0a0e8f464eefcf72025cd445f04981958","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-02-24 14:37:09.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"032bef62a408dbb00b8a4c716fb648746339b123","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-02-25 02:08:01.000000000","message":"Patch Set 11: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/11","accounts_in_message":[],"_revision_number":11},{"id":"6b42689fd48d300c3956cd0c877a977a0e1f1490","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2015-02-25 18:55:29.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d089d8406962c674dce5056d0e8b70bd389cb4c6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-02-27 22:57:54.000000000","message":"Patch Set 11:\n\nI basically tried something similar here but wasn\u0027t sure about rpc compat problems, so this is probably better.\n\nhttps://review.openstack.org/#/c/160033/","accounts_in_message":[],"_revision_number":11},{"id":"6f0450cb7726446c3ae112a7735e68367b516257","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-04 20:01:06.000000000","message":"Patch Set 11: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"70447c8a6141e67e084a49b68380f841b35f9e8a","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-04 20:58:14.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"be722b6e9d9531299c8a9db5994cb468307e1834","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-04 21:23:57.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a02e3747980a40e8b627d36fcc1f6cb4/index.html : SUCCESS in 20m 33s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0a573a4d251f4b438c83560484192271/index.html : SUCCESS in 24m 17s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0efba59a15cf41459987d8478b0f7e74/index.html : SUCCESS in 21m 28s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/12/check/real-db-upgrade_nova_percona_user_001:th-percona/61edfa2e50f04561946497615827f098/index.html : SUCCESS in 24m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/bb1e4b264acd43489662be553bbed3e0/index.html : SUCCESS in 20m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/12/check/real-db-upgrade_nova_percona_user_002:th-percona/11e974f9ad224a36ae077c1b94659a1a/index.html : SUCCESS in 17m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"57365d4748dada3641344c6b0bbb6e08c2acd67b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-04 21:24:04.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"51c88fba85e481f448e16b1e989a1b75459cee46","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-04 21:46:35.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/12/check-ibm-tempest-dsvm-full/6852633 : SUCCESS in 47m 47s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/12/check-ibm-tempest-dsvm-postgres-full/8726b40 : SUCCESS in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4cebee23580e846e72db54f67297b18095c6da54","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-04 22:55:05.000000000","message":"Patch Set 12: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/12/11788/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":12},{"id":"d6103c81be924d85d94b21ea8974797ead3eb45b","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-04 23:49:39.000000000","message":"Patch Set 12:\n\nBuild Successful. Logs at http://198.175.100.33/133664/12","accounts_in_message":[],"_revision_number":12},{"id":"b0d31dccd2652c9e21163d90997cdcc7fd3dd699","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-05 10:47:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"91f98b2911e5ff8a25b46ffa80f99c2ca74a57eb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-05 11:37:22.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/13/check-ibm-tempest-dsvm-full/44052ca : SUCCESS in 49m 14s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/13/check-ibm-tempest-dsvm-postgres-full/b205904 : SUCCESS in 49m 31s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c4d7ce3cc42819c4c21b3fe8077d4ce13fe2bf28","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-05 11:56:29.000000000","message":"Patch Set 13:\n\nBuild Successful. Logs at http://198.175.100.33/133664/13","accounts_in_message":[],"_revision_number":13},{"id":"14741c28ad6f9cb69c7f2ae563672e3f6aa2a0bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-05 11:56:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/13/check/gate-nova-pep8/29b016d/ : SUCCESS in 9m 09s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/13/check/gate-nova-docs/28b11f8//doc/build/html/ : SUCCESS in 9m 03s\n- gate-nova-python27 http://logs.openstack.org/64/133664/13/check/gate-nova-python27/9f4f840/ : SUCCESS in 11m 14s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-full/96c5965/ : SUCCESS in 44m 16s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-postgres-full/7d00efe/ : SUCCESS in 45m 59s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-neutron-full/dfa937c/ : SUCCESS in 1h 02m 05s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/13/check/check-grenade-dsvm/589a846/ : SUCCESS in 44m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/13/check/gate-tempest-dsvm-large-ops/63b64c9/ : SUCCESS in 31m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/13/check/gate-tempest-dsvm-neutron-large-ops/e9258d8/ : SUCCESS in 29m 58s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/13/check/check-devstack-dsvm-cells/124180a/ : SUCCESS in 23m 34s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/13/check/gate-nova-tox-functional/25a24ce/ : SUCCESS in 13m 48s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/13/check/check-grenade-dsvm-partial-ncpu/dd7c23e/ : SUCCESS in 46m 26s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-ironic-pxe_ssh/c5aaae9/ : SUCCESS in 44m 45s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-nova-v21-full/2ef373d/ : SUCCESS in 49m 12s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-cells/79e204e/ : FAILURE in 38m 44s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/13/check/check-tempest-dsvm-full-ceph/f51b209/ : SUCCESS in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ec4dbcf5bdb88ed3d0775d42261adf46e74765c9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-05 11:58:15.000000000","message":"Patch Set 13: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/13/11889/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":13},{"id":"cf143fa2d80f5d3c1cab2f3d37a5572b82b743a7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2015-03-05 15:20:10.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"564ce1701b0ae298c54643b4f7e9d1fbecccae75","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-05 19:14:58.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"585231b047dece876bef30dd57c4bfe372544953","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-05 20:01:01.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/14/check-ibm-tempest-dsvm-full/c4df347 : SUCCESS in 44m 06s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/14/check-ibm-tempest-dsvm-postgres-full/848d6e2 : SUCCESS in 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"59e633e482f0310fea31ab4bfaaaba33cbf95598","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-05 20:01:05.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"d8d70382b7ac89129e94ba338ab45a164c5c71d5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-05 20:32:52.000000000","message":"Patch Set 14: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/14/11966/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":14},{"id":"7d49c10f6c8269cf699383288e5f01ee4ad8916a","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-05 20:42:41.000000000","message":"Patch Set 14:\n\nBuild Successful. Logs at http://198.175.100.33/133664/14","accounts_in_message":[],"_revision_number":14},{"id":"a93e1b26ceaa35cf986860042932374da4f6b2b9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-06 06:58:03.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1d32f3ae1d2342d58d9f3d8a39c7aa68/index.html : SUCCESS in 20m 01s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2e85f39306234a2691474b7ffb627c65/index.html : SUCCESS in 24m 37s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/606dd4db83a646de97b0cc1577a519d7/index.html : SUCCESS in 17m 33s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/14/check/real-db-upgrade_nova_percona_user_001:th-percona/58725219379f4d8ebb2dd468fb6b1f06/index.html : SUCCESS in 24m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/14/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4b2bf9279fdb426a825e96a8e31b87f9/index.html : SUCCESS in 21m 01s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/14/check/real-db-upgrade_nova_percona_user_002:th-percona/b9c93c49cfcf41bf8c2c6a92fb071b14/index.html : SUCCESS in 17m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"44e10ab7b48c6dd970267646902dc752783cbd04","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-06 13:01:34.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"7e4d595c64b593e2639b2b35ac9f7b9a8627ac5e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-06 13:36:38.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/15/check-ibm-tempest-dsvm-full/13068c3 : SUCCESS in 32m 10s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/15/check-ibm-tempest-dsvm-postgres-full/2aa0d32 : SUCCESS in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b6ed6e17db4e0fd9f65cc44dbdf95d5b842b57b1","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-06 13:46:46.000000000","message":"Patch Set 15:\n\nBuild Failed. Logs at http://198.175.100.33/133664/15","accounts_in_message":[],"_revision_number":15},{"id":"67d83b0a5a1f0157c58bd2c4951a2a69dd490105","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-06 14:01:01.000000000","message":"Patch Set 15: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/15/12111/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":15},{"id":"c8a69f7545a7ed3aa80b13470048a77697fa0427","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-06 14:21:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/15/check/gate-nova-pep8/de72730/ : SUCCESS in 8m 07s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/15/check/gate-nova-docs/be1c8cd//doc/build/html/ : SUCCESS in 10m 35s\n- gate-nova-python27 http://logs.openstack.org/64/133664/15/check/gate-nova-python27/efddf3a/ : SUCCESS in 12m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-full/1c57c06/ : SUCCESS in 48m 13s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-postgres-full/428bbb3/ : SUCCESS in 59m 30s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-neutron-full/fde5c01/ : SUCCESS in 1h 16m 01s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/15/check/check-grenade-dsvm/76206e4/ : SUCCESS in 39m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/15/check/gate-tempest-dsvm-large-ops/31b8fbf/ : SUCCESS in 26m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/15/check/gate-tempest-dsvm-neutron-large-ops/168344f/ : SUCCESS in 30m 08s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/15/check/check-devstack-dsvm-cells/b6a656f/ : SUCCESS in 24m 55s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/15/check/gate-nova-tox-functional/a02660d/ : SUCCESS in 15m 22s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/15/check/check-grenade-dsvm-partial-ncpu/e1a87b0/ : SUCCESS in 52m 20s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-ironic-pxe_ssh/12e64b7/ : SUCCESS in 42m 02s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-nova-v21-full/5aadfe2/ : SUCCESS in 48m 43s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-cells/f2e78a8/ : FAILURE in 38m 51s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/15/check/check-tempest-dsvm-full-ceph/97434e0/ : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"816519f1717510fc980aae1a0bccac774fc9f924","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-06 15:27:17.000000000","message":"Patch Set 15: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/15/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2633a2b737d441c2875bae5e200f9646/index.html : SUCCESS in 18m 35s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/15/check/real-db-upgrade_nova_mysql_user_001:th-mysql/34a7683b36b343398d09a2b0bd5c590d/index.html : SUCCESS in 24m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/15/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/442965b7df494830984eeaf154a113dc/index.html : SUCCESS in 20m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/15/check/real-db-upgrade_nova_percona_user_001:th-percona/06ba69ea040241b9bfd685d08d7427b9/index.html : SUCCESS in 24m 03s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/15/check/real-db-upgrade_nova_mysql_user_002:th-mysql/44cca2c6a41240659d63f15342b45735/index.html : SUCCESS in 21m 01s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/15/check/real-db-upgrade_nova_percona_user_002:th-percona/cab920fcd6c448abaf439fa6033950e0/index.html : SUCCESS in 18m 08s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":15},{"id":"f63cc25edb4675b01f6fd259b3f0f1e7a0f36a7b","author":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"date":"2015-03-06 16:24:25.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"3caf0999e8c86f95f7760f3036dc7a7c049dbf61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-09 00:32:02.000000000","message":"Patch Set 15: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/15","accounts_in_message":[],"_revision_number":15},{"id":"5762629edd41c0f2d583fe287ee3e7ad09584eff","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-03-10 23:11:50.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"680749a16a50fc1287a2fabbe04edaa883c02bc9","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2015-03-11 09:39:20.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"5ee6ab0286e4ede6ae5323b94f08910dd43c291b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-12 18:35:59.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"30e1ba7d0b1bd4d80dec28516d727b63b077be13","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-13 11:48:37.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"24d3f93179074a4ad0af47e6b4474f63120a38d6","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-13 12:11:19.000000000","message":"Patch Set 16: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0a3bf261f3f9497bbb7bc39b517fa2e3/index.html : SUCCESS in 21m 31s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a1361b4cbd8c43c69c899c920dceb7c1/index.html : SUCCESS in 19m 42s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/56c3cfa4f1ce40baaa10565095f69762/index.html : SUCCESS in 17m 05s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_percona_user_001:th-percona/45824854229b4683a513fe8fb1b2141d/index.html : SUCCESS in 19m 45s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0fbbf8aa3a724168a91bcade3f481a89/index.html : SUCCESS in 20m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_percona_user_002:th-percona/dbad095f8cc645c1a01de27a30f37282/index.html : SUCCESS in 17m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":16},{"id":"5a5d29bbb71869223ef2267266e2cfcf3b8cf96e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-13 12:18:16.000000000","message":"Patch Set 16:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/16/check-ibm-tempest-dsvm-full/1a67076 : SUCCESS in 28m 47s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/16/check-ibm-tempest-dsvm-postgres-full/174ab87 : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"347aa17cd071e7e6f126df45ac44b1cc3ef1e82f","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-13 12:37:48.000000000","message":"Patch Set 16:\n\nBuild Failed. Logs at http://198.175.100.33/133664/16","accounts_in_message":[],"_revision_number":16},{"id":"1c97254807fa42726e2607ebf2ff0ecc2dc4e7b0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-13 12:49:44.000000000","message":"Patch Set 16: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/16/13134/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":16},{"id":"6d471b309ec3e72c3c9d828e822ca61d99d8125f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-13 13:24:42.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/16/check/gate-nova-pep8/b81200f/ : SUCCESS in 7m 57s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/16/check/gate-nova-docs/b7233fb//doc/build/html/ : SUCCESS in 8m 58s\n- gate-nova-python27 http://logs.openstack.org/64/133664/16/check/gate-nova-python27/d762f51/ : SUCCESS in 13m 13s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-full/6a29325/ : SUCCESS in 42m 33s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-postgres-full/8a995fd/ : SUCCESS in 39m 45s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-neutron-full/c61eabb/ : SUCCESS in 1h 00m 08s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/16/check/check-grenade-dsvm/716baa2/ : SUCCESS in 33m 39s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/16/check/gate-tempest-dsvm-large-ops/f948544/ : SUCCESS in 32m 36s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/16/check/gate-tempest-dsvm-neutron-large-ops/720204a/ : FAILURE in 31m 02s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/16/check/check-devstack-dsvm-cells/67020cf/ : SUCCESS in 19m 55s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/16/check/gate-nova-tox-functional/c9e3c9b/ : SUCCESS in 10m 54s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/16/check/check-grenade-dsvm-partial-ncpu/6f2dc87/ : SUCCESS in 42m 32s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-ironic-pxe_ssh/b7ed3d8/ : SUCCESS in 37m 37s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-nova-v21-full/3edc2cf/ : SUCCESS in 45m 37s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-cells/27789cc/ : FAILURE in 32m 54s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-full-ceph/b70533c/ : SUCCESS in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"2ab1db41e6676c874ed88715c4428e544f7b93b6","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-13 13:32:23.000000000","message":"Patch Set 16:\n\nrecheck bug 1403284","accounts_in_message":[],"_revision_number":16},{"id":"269e8a053250913fac3bff6e9e0fbd2e36cc3613","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-13 13:57:14.000000000","message":"Patch Set 16:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/64939033d95e41eda1283e6e313326c8/index.html : SUCCESS in 20m 17s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f1ff97aac9734524a6b6e8f284bdf2fd/index.html : SUCCESS in 20m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d482a461a53147e0b9f09032f91e7abf/index.html : SUCCESS in 16m 55s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_percona_user_001:th-percona/3810d191f7b6494ea431b99ad0534912/index.html : SUCCESS in 24m 01s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ea85b95b4db84f2c9442b2ea6ce587fb/index.html : SUCCESS in 21m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/16/check/real-db-upgrade_nova_percona_user_002:th-percona/79776088c0284056b0c541ac463450ae/index.html : SUCCESS in 21m 25s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":16},{"id":"69d45e02d3b84e219d724a0d023826507b584a54","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-13 15:58:42.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/16/check/gate-nova-pep8/94c4e0f/ : SUCCESS in 8m 09s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/16/check/gate-nova-docs/ab96a1d//doc/build/html/ : SUCCESS in 10m 27s\n- gate-nova-python27 http://logs.openstack.org/64/133664/16/check/gate-nova-python27/c64f0e3/ : SUCCESS in 17m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-full/46d7f0f/ : SUCCESS in 45m 25s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-postgres-full/283b830/ : SUCCESS in 39m 43s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-neutron-full/48eb812/ : SUCCESS in 56m 09s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/16/check/check-grenade-dsvm/a1f45bf/ : SUCCESS in 47m 07s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/16/check/gate-tempest-dsvm-large-ops/91e717e/ : SUCCESS in 28m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/16/check/gate-tempest-dsvm-neutron-large-ops/82a5b2a/ : SUCCESS in 26m 25s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/16/check/check-devstack-dsvm-cells/9d9dd22/ : SUCCESS in 20m 31s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/16/check/gate-nova-tox-functional/d75a23a/ : SUCCESS in 14m 58s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/16/check/check-grenade-dsvm-partial-ncpu/e3d6e47/ : SUCCESS in 43m 19s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-ironic-pxe_ssh/c22da64/ : SUCCESS in 38m 33s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-nova-v21-full/2479cfa/ : SUCCESS in 40m 08s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-cells/2c2b7a5/ : FAILURE in 33m 37s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/16/check/check-tempest-dsvm-full-ceph/634f967/ : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"52004f13d5b3fb75d630f5f9734acbfdb53f92e0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-13 18:31:10.000000000","message":"Patch Set 16: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"d016b800ab446b271a67189d847f3ed08448661b","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-13 19:40:47.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"24f1b157a1a65a95b6b77710f124fc8d1cbc340f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-13 20:10:19.000000000","message":"Patch Set 17: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/17/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5b84e37451c64bb3bef362c4c30fe3fd/index.html : SUCCESS in 21m 16s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/17/check/real-db-upgrade_nova_mysql_user_001:th-mysql/04906ad93ce844ee96b4e371736076c1/index.html : SUCCESS in 23m 22s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/17/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/be4a693d6de1430a8f9bba94dc396174/index.html : SUCCESS in 20m 30s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/17/check/real-db-upgrade_nova_percona_user_001:th-percona/f74e3eff1b4d417795de7ca92f097f6c/index.html : SUCCESS in 24m 03s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/17/check/real-db-upgrade_nova_mysql_user_002:th-mysql/17316c4319b1480cb164165b8eab494d/index.html : SUCCESS in 17m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/17/check/real-db-upgrade_nova_percona_user_002:th-percona/8956a08a50334aa2981a4bf3f7e1971c/index.html : SUCCESS in 17m 56s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":17},{"id":"2867b3909aede4db995bb9530bfaf0d9e5d236f4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-13 20:10:28.000000000","message":"Patch Set 17:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"a6f0ccb974b7b09bd1e42b82e5ec8657a1c80e03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-13 20:10:31.000000000","message":"Patch Set 17: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"42d4ca239d10522b795be8f8e1e0133392ccf653","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-13 20:44:23.000000000","message":"Patch Set 17: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/17/13236/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":17},{"id":"ed9d2e5b8be924e3a0d1b7c995c66d8073b27f62","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-13 22:48:12.000000000","message":"Patch Set 17:\n\nBuild Successful. Logs at http://198.175.100.33/133664/17","accounts_in_message":[],"_revision_number":17},{"id":"992a0922e4e71031c5620dc871ac4f80db63801b","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-15 08:41:59.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"22c12dd4e84525cdf4b54eb330eb3cf1e017f4c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-15 09:50:26.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/17/check/gate-nova-pep8/e56f33e/ : SUCCESS in 8m 42s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/17/check/gate-nova-docs/37663ff//doc/build/html/ : SUCCESS in 9m 10s\n- gate-nova-python27 http://logs.openstack.org/64/133664/17/check/gate-nova-python27/6c04468/ : SUCCESS in 17m 15s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-full/79f8c7e/ : SUCCESS in 42m 13s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-postgres-full/4e82cff/ : SUCCESS in 38m 52s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-neutron-full/7b078db/ : SUCCESS in 59m 28s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/17/check/check-grenade-dsvm/6e132e0/ : SUCCESS in 42m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/17/check/gate-tempest-dsvm-large-ops/a4f1d31/ : SUCCESS in 27m 25s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/17/check/gate-tempest-dsvm-neutron-large-ops/440cbbc/ : SUCCESS in 25m 14s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/17/check/check-devstack-dsvm-cells/5cc77e4/ : SUCCESS in 18m 52s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/17/check/gate-nova-tox-functional/af0d318/ : SUCCESS in 14m 34s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/17/check/check-grenade-dsvm-partial-ncpu/e00ff21/ : SUCCESS in 44m 46s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-ironic-pxe_ssh/213882d/ : SUCCESS in 38m 00s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-nova-v21-full/00730ae/ : SUCCESS in 37m 40s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-cells/d6c6bc2/ : FAILURE in 33m 20s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/17/check/check-tempest-dsvm-full-ceph/23af0cb/ : SUCCESS in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"db677a1be08d48be50c37aa097ae2e7b16aac9f3","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2015-03-16 03:41:03.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"a9031b30a1e27d2c3573f9ec2ccd10b30afbf082","author":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"date":"2015-03-16 15:16:50.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"7d7393feab9f1cfbe39791abc68da83be212ac71","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-18 19:18:28.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"3ea9bac6324e5fe87bf2b36ac16785318c4391ed","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-18 19:58:24.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2478c70582131a4967f8b38c702b9857fb8a6e97","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-18 20:20:15.000000000","message":"Patch Set 18:\n\nBuild Successful. Logs at http://198.175.100.33/133664/18","accounts_in_message":[],"_revision_number":18},{"id":"8464f82b6379c6c930180ec7af62d52f38c719bc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-03-18 20:21:34.000000000","message":"Patch Set 18:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/133664/18/ : Test run failed. in 22m 36s","accounts_in_message":[],"_revision_number":18},{"id":"129b17bee5920c4065469512f5e22e981d770b45","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-18 20:24:52.000000000","message":"Patch Set 18: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/18/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6e820438a2c24264a970f1750fbac3ce/index.html : SUCCESS in 21m 49s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/18/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e0fbd64c98cc4e7e9cb05b3b52cddf30/index.html : SUCCESS in 25m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/18/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8b2985e8ab0b4061a245e83025987c5a/index.html : SUCCESS in 18m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/18/check/real-db-upgrade_nova_percona_user_001:th-percona/146beee1904c4ee89729ed9de561b869/index.html : SUCCESS in 24m 51s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/18/check/real-db-upgrade_nova_mysql_user_002:th-mysql/134f7243431345c3b3de544d519783eb/index.html : SUCCESS in 18m 44s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/18/check/real-db-upgrade_nova_percona_user_002:th-percona/8151e999f2e64940a7ebd58a51161d8e/index.html : SUCCESS in 22m 29s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":18},{"id":"03868c81ccfcfac3e768592eb110e86c9436dcd4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-18 21:00:02.000000000","message":"Patch Set 18: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/18/13876/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":18},{"id":"0d09d093ebcaeba837577353a93a207b01efb806","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2015-03-18 21:00:25.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"fbc09d8eae35ca3646d90bc628896c157ad53704","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-18 21:06:31.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/18/check-ibm-tempest-dsvm-full/54f81e7 : SUCCESS in 1h 07m 15s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/18/check-ibm-tempest-dsvm-postgres-full/1c0d26c : SUCCESS in 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2e641dc99cd1ff14a3b8d220042ba0bf1e7fd86a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-19 00:15:56.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/18/check/gate-nova-pep8/ad92f21/ : SUCCESS in 8m 35s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/18/check/gate-nova-docs/05a71af//doc/build/html/ : SUCCESS in 8m 51s\n- gate-nova-python27 http://logs.openstack.org/64/133664/18/check/gate-nova-python27/83e8447/ : SUCCESS in 12m 58s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-full/413ab08/ : SUCCESS in 38m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-postgres-full/58dc3d2/ : SUCCESS in 1h 21m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-neutron-full/5b36039/ : SUCCESS in 53m 21s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/18/check/check-grenade-dsvm/a4f4521/ : SUCCESS in 43m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/18/check/gate-tempest-dsvm-large-ops/e1727e1/ : SUCCESS in 24m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/18/check/gate-tempest-dsvm-neutron-large-ops/001c098/ : SUCCESS in 25m 55s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/18/check/check-devstack-dsvm-cells/b479550/ : SUCCESS in 26m 47s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/18/check/gate-nova-tox-functional/f255670/ : SUCCESS in 11m 20s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/18/check/check-grenade-dsvm-partial-ncpu/1aea2f9/ : SUCCESS in 50m 40s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-ironic-pxe_ssh/fea07b3/ : SUCCESS in 36m 26s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-nova-v21-full/c3f72cc/ : SUCCESS in 44m 14s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-cells/938dc77/ : FAILURE in 32m 15s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/18/check/check-tempest-dsvm-full-ceph/ca0bfad/ : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"cee1c6180937d74c9661089aedd1d1494b137d8a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-20 06:03:32.000000000","message":"Patch Set 18: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/18","accounts_in_message":[],"_revision_number":18},{"id":"c4dd060ce6afa070ff0d367df846d0eb6240192d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-20 06:04:28.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":18},{"id":"ac58bed56a4725743f471cb41f77448e1ad74b53","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-20 08:32:09.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"68a6fbb44d6f48de9f0fab2329031ff4d29d85f0","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-20 08:34:59.000000000","message":"Patch Set 19:\n\nrebase","accounts_in_message":[],"_revision_number":19},{"id":"271695059902230ccb90e1029d6ce873371bfa03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-20 08:35:02.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"bbce6d31340a787f128f05bcfe13c9ef09e5423e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-20 08:56:43.000000000","message":"Patch Set 19: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/19/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6992c8aebd28444890358cd323c6f3bb/index.html : SUCCESS in 20m 51s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/19/check/real-db-upgrade_nova_mysql_user_001:th-mysql/811199793bf6475c81efe51f5cfdefcf/index.html : SUCCESS in 21m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/19/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4683d4458f28477c96ab305a6e2171af/index.html : SUCCESS in 17m 44s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/19/check/real-db-upgrade_nova_percona_user_001:th-percona/2ec0e99b1c484cd8a54004b2761467d4/index.html : SUCCESS in 20m 33s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/19/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e1911a4093524c6ebf2096b746e65c64/index.html : SUCCESS in 18m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/19/check/real-db-upgrade_nova_percona_user_002:th-percona/828c8e09bb144d81bfdf91073a5aea09/index.html : SUCCESS in 23m 31s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":19},{"id":"7a156aa394e0a14a8d71c05a170b173956abd80f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-20 09:30:10.000000000","message":"Patch Set 19: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/19/14096/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":19},{"id":"ee820c507e1d03c7436ee73fdbe6963d969a030b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-20 09:32:31.000000000","message":"Patch Set 19:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/19/check-ibm-tempest-dsvm-full/54a299f : SUCCESS in 59m 31s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/19/check-ibm-tempest-dsvm-postgres-full/950f512 : SUCCESS in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"3187c964565532fea7058865ec35d9b4ffc9fe97","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-20 10:20:48.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"bdfb4a066347e140d66cd9374ad8c123f6b4eaf4","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-20 10:46:55.000000000","message":"Patch Set 19:\n\nBuild Successful. Logs at http://198.175.100.33/133664/19","accounts_in_message":[],"_revision_number":19},{"id":"d6cbc3cdf7a967bacc997c5aaf9620bc60419c60","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-03-20 10:58:13.000000000","message":"Patch Set 19: Code-Review+1\n\nIMHO, the change is okay, you can not bump a new version and leave the old methods deprecated if a old client is asking for them.","accounts_in_message":[],"_revision_number":19},{"id":"33be9a022804905c533c8c552fd2846f4c084b88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-20 11:56:42.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/19/check/gate-nova-pep8/592f295/ : SUCCESS in 8m 46s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/19/check/gate-nova-docs/17cd8f7//doc/build/html/ : SUCCESS in 8m 17s\n- gate-nova-python27 http://logs.openstack.org/64/133664/19/check/gate-nova-python27/50ce0de/ : SUCCESS in 14m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-full/4a44542/ : SUCCESS in 38m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-postgres-full/5f90762/ : SUCCESS in 38m 58s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-neutron-full/76e686b/ : SUCCESS in 50m 04s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/19/check/check-grenade-dsvm/51e9013/ : SUCCESS in 39m 53s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/19/check/gate-tempest-dsvm-large-ops/be5a114/ : SUCCESS in 22m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/19/check/gate-tempest-dsvm-neutron-large-ops/5e21319/ : SUCCESS in 22m 16s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/19/check/check-devstack-dsvm-cells/ab6c7b8/ : SUCCESS in 21m 01s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/19/check/gate-nova-tox-functional/761bcd6/ : SUCCESS in 12m 24s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/19/check/check-grenade-dsvm-partial-ncpu/3a258eb/ : SUCCESS in 44m 19s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-ironic-pxe_ssh/4182e6a/ : SUCCESS in 37m 44s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-nova-v21-full/9df9758/ : SUCCESS in 37m 50s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-cells/c3880a7/ : FAILURE in 27m 37s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/19/check/check-tempest-dsvm-full-ceph/69a511e/ : SUCCESS in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"1bc925ab4e2edc93842403e2d140c8c1441089cc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-23 16:32:31.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"cbf70679a1a174a475af38116a5af5b90311da9d","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-23 17:58:12.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"96025d41d7262149d418046f0e161c90edda36ec","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-23 18:00:33.000000000","message":"Patch Set 20:\n\nrebase","accounts_in_message":[],"_revision_number":20},{"id":"2fd0ffec6c6368c3a3189546001a1917a7618606","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-23 18:00:35.000000000","message":"Patch Set 20: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":20},{"id":"839f4581a52d8439ca49211e66a08e795bc00416","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-23 18:15:23.000000000","message":"Patch Set 20:\n\nBuild Successful. Logs at http://198.175.100.33/133664/20","accounts_in_message":[],"_revision_number":20},{"id":"957c92a4ba6f24f92286ed1aa7869c57e2df03b0","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-03-23 18:21:09.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"3540ad64b181d9f3d796a8839cccf37f2b274b7f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-23 18:24:37.000000000","message":"Patch Set 20: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/20/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1a458824d7f949dd9a22a3dfc5506166/index.html : SUCCESS in 22m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/20/check/real-db-upgrade_nova_mysql_user_001:th-mysql/805eab3deda141259c5a753a3c32d3b8/index.html : SUCCESS in 20m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/20/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8658bb54a46948bb9eed4e223c12e929/index.html : SUCCESS in 22m 12s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/20/check/real-db-upgrade_nova_percona_user_001:th-percona/a3d725c7a3d242bf8545ba44bbd7804b/index.html : SUCCESS in 25m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/20/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c662c2117cf949fa9bba9536f58b38f4/index.html : SUCCESS in 21m 35s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/20/check/real-db-upgrade_nova_percona_user_002:th-percona/d7cccd4eb197473db927530af9b01c6f/index.html : SUCCESS in 18m 00s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":20},{"id":"848be22626ad5e4947a00888980c720553743b0b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-23 19:00:13.000000000","message":"Patch Set 20:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/20/check-ibm-tempest-dsvm-full/c230000 : SUCCESS in 1h 01m 11s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/20/check-ibm-tempest-dsvm-postgres-full/ae0ad6b : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"675ca717440db3568534e778693925668bca72f0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-23 19:01:25.000000000","message":"Patch Set 20: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/64/133664/20/14241/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":20},{"id":"0ff0c7bcc8058156a33828d58c8275fa7263e151","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-23 19:23:00.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/20/check/gate-nova-pep8/3704257/ : SUCCESS in 11m 10s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/20/check/gate-nova-docs/9200a0f//doc/build/html/ : SUCCESS in 11m 20s\n- gate-nova-python27 http://logs.openstack.org/64/133664/20/check/gate-nova-python27/87d6d6f/ : SUCCESS in 13m 40s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-full/a6f2126/ : SUCCESS in 42m 01s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-postgres-full/90b0728/ : SUCCESS in 44m 12s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-neutron-full/15a48ae/ : SUCCESS in 56m 22s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/20/check/check-grenade-dsvm/f80894f/ : SUCCESS in 38m 42s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/20/check/gate-tempest-dsvm-large-ops/cdb8944/ : SUCCESS in 27m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/20/check/gate-tempest-dsvm-neutron-large-ops/ba9adff/ : SUCCESS in 24m 50s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/20/check/check-devstack-dsvm-cells/8291477/ : SUCCESS in 19m 33s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/20/check/gate-nova-tox-functional/1d062ce/ : SUCCESS in 12m 24s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/20/check/check-grenade-dsvm-partial-ncpu/ac2b2be/ : SUCCESS in 35m 33s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-ironic-pxe_ssh/dc90b99/ : SUCCESS in 32m 37s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-nova-v21-full/bf30f86/ : SUCCESS in 38m 55s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-cells/f6c4999/ : FAILURE in 30m 18s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/20/check/check-tempest-dsvm-full-ceph/3a331ba/ : SUCCESS in 35m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"891f7f63d8696fc7f562074cab80252296bd62a1","author":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"date":"2015-03-23 21:12:34.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"9fd0f5132821d3476b1f21ae774a9130d52f8424","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2015-03-23 23:38:06.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"bb9c158016a8e0b58784bcdf62c7eb9cee61388e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-25 11:31:43.000000000","message":"Patch Set 19: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/19","accounts_in_message":[],"_revision_number":19},{"id":"22bf67397a1e60802c795bb51131d879c2b7257f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-03-25 13:33:09.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"5ab1eb7b5e9aeed4188662084ab703d5ec59d525","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2015-03-26 07:25:20.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"4ea8c27f9a233a1d0a8abfbc1fab6d4e77d2edf7","author":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"date":"2015-03-26 10:51:53.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"b57bd2817fee3573167b3bf41f6c7d9b30780769","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-28 12:40:16.000000000","message":"Patch Set 20: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/133664/20 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/133664/20 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"2070b0ab12af9f665a3a60cced7953c02294f6ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-13 16:23:55.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"2ecc42106378c1b5c5aa8f58ab93ec7c9b2c1ca2","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-04-13 17:49:42.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"ff1286b0e2f1438cda90a2848bf2965b8457e02c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-04-13 18:03:30.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"717de007936167a546a2241c9732dfcbff49a6cb","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-13 18:04:28.000000000","message":"Patch Set 21:\n\n* pci-test http://198.175.100.33/133664/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"f8b4f33fc21a0703b136d0c44228184f09b17823","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-13 18:17:04.000000000","message":"Patch Set 21: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/21/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/47f938d094524f8283f0efb8ff540120/index.html : SUCCESS in 18m 59s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/21/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d27ddbce92c04b44a11d7251ca3fe9aa/index.html : SUCCESS in 22m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/21/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/aa7d2f93dcea44bdaa1edb4dbb7c7a9d/index.html : SUCCESS in 18m 45s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/21/check/real-db-upgrade_nova_percona_user_001:th-percona/5143566c70b5414292e72da993465044/index.html : SUCCESS in 26m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/64/133664/21/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a4c4519bd7324efbb3e8c19346bf73f7/index.html : SUCCESS in 23m 12s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/64/133664/21/check/real-db-upgrade_nova_percona_user_002:th-percona/b360548254924c7885d5e967daed7dd6/index.html : SUCCESS in 23m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":21},{"id":"140933294033c2a7f24efb6fce5107904d443ca0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-13 19:05:53.000000000","message":"Patch Set 21:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/21/check-ibm-tempest-dsvm-full/3d97740 : SUCCESS in 1h 11m 21s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/133664/21/check-ibm-tempest-dsvm-postgres-full/3433a7e : SUCCESS in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"90983bf783e2f5f89361a639a9f9f392f0e7612c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-13 19:08:27.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/64/133664/21/check/gate-nova-pep8/e8a49fd/ : SUCCESS in 8m 14s\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/21/check/gate-nova-docs/bb9d413//doc/build/html/ : SUCCESS in 9m 04s\n- gate-nova-python27 http://logs.openstack.org/64/133664/21/check/gate-nova-python27/b05b2b4/ : SUCCESS in 19m 54s\n- check-tempest-dsvm-full http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-full/3319d38/ : SUCCESS in 42m 36s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-postgres-full/8f65598/ : SUCCESS in 44m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-neutron-full/e0d1e9e/ : SUCCESS in 1h 01m 23s\n- check-grenade-dsvm http://logs.openstack.org/64/133664/21/check/check-grenade-dsvm/a99a06f/ : SUCCESS in 40m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/21/check/gate-tempest-dsvm-large-ops/bfc974a/ : SUCCESS in 24m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/21/check/gate-tempest-dsvm-neutron-large-ops/4f2c024/ : SUCCESS in 20m 16s\n- check-devstack-dsvm-cells http://logs.openstack.org/64/133664/21/check/check-devstack-dsvm-cells/b2f81e1/ : SUCCESS in 19m 08s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/21/check/gate-nova-tox-functional/fc815d8/ : SUCCESS in 10m 47s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/21/check/check-grenade-dsvm-partial-ncpu/d5a013d/ : SUCCESS in 42m 05s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-ironic-pxe_ssh/92e48ff/ : SUCCESS in 34m 56s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-nova-v21-full/340975d/ : SUCCESS in 49m 10s\n- check-tempest-dsvm-cells http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-cells/bb80aae/ : FAILURE in 39m 40s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-full-ceph/6b63cf6/ : SUCCESS in 41m 42s\n- check-tempest-dsvm-aiopcpu-full http://logs.openstack.org/64/133664/21/check/check-tempest-dsvm-aiopcpu-full/8959947/ : FAILURE in 49m 21s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"36e627200a4e9ac34aa808d2378e8f871f2849b1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-04-13 19:09:23.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/133664/21/ : FAILURE in 1h 19m 06s","accounts_in_message":[],"_revision_number":21},{"id":"cbd2d90ca5ab1ea0f477594116cff21b003a251d","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2015-04-14 04:54:31.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"78fb172362ea98fc0a860ed13ea48325fc180a97","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2015-04-14 16:19:48.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"3cde0f80b31c670ce150eaf9bd542a72dfa152c8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2015-04-14 22:09:06.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"0b2075cc00fb0a04719b31fecd985f4a8436af63","author":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"date":"2015-04-15 16:38:19.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"6ddbe7743b12e6711de2fba748777bb2eb02fbe9","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2015-04-17 16:41:11.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"fe6984cd36c55c30f8d4a9a57ee3ed655686471e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-17 16:41:18.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":21},{"id":"2d29e265c093a7f2d4cdf4a48ec3853cc981e210","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-17 18:02:50.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/64/133664/21/gate/gate-nova-docs/caf2163//doc/build/html/ : SUCCESS in 8m 20s\n- gate-nova-pep8 http://logs.openstack.org/64/133664/21/gate/gate-nova-pep8/2a6525e/ : SUCCESS in 8m 23s\n- gate-nova-python27 http://logs.openstack.org/64/133664/21/gate/gate-nova-python27/e33e20f/ : SUCCESS in 14m 18s\n- gate-tempest-dsvm-full http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-full/e98a91c/ : SUCCESS in 41m 32s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-postgres-full/8db63c0/ : SUCCESS in 42m 58s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-neutron-full/fc340ef/ : SUCCESS in 1h 06m 58s\n- gate-grenade-dsvm http://logs.openstack.org/64/133664/21/gate/gate-grenade-dsvm/8e0c603/ : SUCCESS in 42m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-large-ops/a0e0e1d/ : SUCCESS in 24m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-neutron-large-ops/73e60df/ : SUCCESS in 25m 24s\n- gate-nova-tox-functional http://logs.openstack.org/64/133664/21/gate/gate-nova-tox-functional/4d06791/ : SUCCESS in 12m 48s\n- gate-devstack-dsvm-cells http://logs.openstack.org/64/133664/21/gate/gate-devstack-dsvm-cells/a828ecd/ : SUCCESS in 20m 59s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/64/133664/21/gate/gate-grenade-dsvm-partial-ncpu/181cdb9/ : SUCCESS in 45m 21s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-nova-v21-full/06aa581/ : SUCCESS in 45m 14s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/64/133664/21/gate/gate-tempest-dsvm-full-ceph/23a3c08/ : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":21},{"id":"c6f27fdbc9934e7c050128612ebaf94086756811","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-17 18:02:53.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"6285d6a5114b739697f8d0041e31b0b945facf7c","revisions":{"d782c4c4bdcb2cc6aa8a4ed47b6a9f5f8047ca06":{"kind":"REWORK","_number":1,"created":"2014-11-11 12:10:42.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/1"}}},"commit":{"parents":[{"commit":"fb3482c48ba9c0177166831e74ad3458e11a84ff","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb3482c48ba9c0177166831e74ad3458e11a84ff"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-11 12:10:41.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d782c4c4bdcb2cc6aa8a4ed47b6a9f5f8047ca06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d782c4c4bdcb2cc6aa8a4ed47b6a9f5f8047ca06"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fb3482c48ba9c0177166831e74ad3458e11a84ff","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cce687da1210c646e8b107e925993dc9dd5b0b6c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2014-11-11 12:34:20.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/2"}}},"commit":{"parents":[{"commit":"04a1ee2b26f2a1ad87124bdb035fc80db8890912","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04a1ee2b26f2a1ad87124bdb035fc80db8890912"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-11 12:28:55.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cce687da1210c646e8b107e925993dc9dd5b0b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cce687da1210c646e8b107e925993dc9dd5b0b6c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"04a1ee2b26f2a1ad87124bdb035fc80db8890912","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a0d149e78f084d102998ec1e5f36bb1a202fb104":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2014-11-11 14:13:13.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/3"}}},"commit":{"parents":[{"commit":"ed374c819531a40dd6efdb92ea25a90fcb916661","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed374c819531a40dd6efdb92ea25a90fcb916661"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-11 14:07:32.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0d149e78f084d102998ec1e5f36bb1a202fb104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0d149e78f084d102998ec1e5f36bb1a202fb104"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ed374c819531a40dd6efdb92ea25a90fcb916661","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a0763ccd25cd2f2f92841beec7771a89d12ab4d9":{"kind":"REWORK","_number":4,"created":"2014-11-13 08:15:37.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/4"}}},"commit":{"parents":[{"commit":"1fa4d3c22396e971c41e11dccd121b3717aff9fa","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fa4d3c22396e971c41e11dccd121b3717aff9fa"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-13 08:08:37.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0763ccd25cd2f2f92841beec7771a89d12ab4d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0763ccd25cd2f2f92841beec7771a89d12ab4d9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1fa4d3c22396e971c41e11dccd121b3717aff9fa","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8b4d129359a1560796101cc299e1e14ffa2568f7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2014-11-22 10:10:40.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/5"}}},"commit":{"parents":[{"commit":"32e8b9cff51b994060eaa85684f18d1ef02ed283","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32e8b9cff51b994060eaa85684f18d1ef02ed283"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-22 10:09:26.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b4d129359a1560796101cc299e1e14ffa2568f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b4d129359a1560796101cc299e1e14ffa2568f7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"32e8b9cff51b994060eaa85684f18d1ef02ed283","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b4c930c2686d2b31828af187ae586a36df1475dd":{"kind":"REWORK","_number":6,"created":"2014-12-08 21:17:55.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/6"}}},"commit":{"parents":[{"commit":"17aded08bbe911060f79aedbfc3a1b66e94bc9ae","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17aded08bbe911060f79aedbfc3a1b66e94bc9ae"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-12-08 21:17:27.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4c930c2686d2b31828af187ae586a36df1475dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4c930c2686d2b31828af187ae586a36df1475dd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"17aded08bbe911060f79aedbfc3a1b66e94bc9ae","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0a1cb2d7ad9f1b9a6aea4ad281dac452565432d2":{"kind":"REWORK","_number":7,"created":"2015-01-22 11:54:02.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/7"}}},"commit":{"parents":[{"commit":"f201ed27f8fdeff9de86d3f72ec16c2e3333fdda","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f201ed27f8fdeff9de86d3f72ec16c2e3333fdda"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-01-22 11:43:13.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a1cb2d7ad9f1b9a6aea4ad281dac452565432d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a1cb2d7ad9f1b9a6aea4ad281dac452565432d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f201ed27f8fdeff9de86d3f72ec16c2e3333fdda","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"349cea8ab45402732684c12444d9d474e82c0186":{"kind":"REWORK","_number":8,"created":"2015-01-22 12:00:17.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/8"}}},"commit":{"parents":[{"commit":"f201ed27f8fdeff9de86d3f72ec16c2e3333fdda","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f201ed27f8fdeff9de86d3f72ec16c2e3333fdda"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-01-22 12:00:16.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/349cea8ab45402732684c12444d9d474e82c0186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/349cea8ab45402732684c12444d9d474e82c0186"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f201ed27f8fdeff9de86d3f72ec16c2e3333fdda","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b02c233dcaca58ab88729beea6b31681929da326":{"kind":"REWORK","_number":9,"created":"2015-02-06 17:53:59.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/9"}}},"commit":{"parents":[{"commit":"4a6c3b232b8a8b9a7763481fb8a1da31541826e2","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a6c3b232b8a8b9a7763481fb8a1da31541826e2"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-02-06 17:38:59.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b02c233dcaca58ab88729beea6b31681929da326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b02c233dcaca58ab88729beea6b31681929da326"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4a6c3b232b8a8b9a7763481fb8a1da31541826e2","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bdf60621b17592b9deca94425fa06d5f26b3a0db":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2015-02-18 09:18:31.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/10"}}},"commit":{"parents":[{"commit":"7e572ee6b712771a7efa172bd47a85954be7945c","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e572ee6b712771a7efa172bd47a85954be7945c"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-02-18 09:04:50.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bdf60621b17592b9deca94425fa06d5f26b3a0db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bdf60621b17592b9deca94425fa06d5f26b3a0db"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7e572ee6b712771a7efa172bd47a85954be7945c","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0b47d9659896d3af6afb7e3b04b4ee20a6d69997":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2015-02-23 08:08:25.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/11"}}},"commit":{"parents":[{"commit":"7ee277d670c14da8ca221a6be8620b4db54f5446","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ee277d670c14da8ca221a6be8620b4db54f5446"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-02-23 07:58:26.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b47d9659896d3af6afb7e3b04b4ee20a6d69997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b47d9659896d3af6afb7e3b04b4ee20a6d69997"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ee277d670c14da8ca221a6be8620b4db54f5446","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8f066ef22bc156b3ab525b8202a526c310777bad":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2015-03-04 20:58:14.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/12"}}},"commit":{"parents":[{"commit":"c3b7c58a9ed05f98e290ec26019fb46f277ebd68","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3b7c58a9ed05f98e290ec26019fb46f277ebd68"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-04 20:48:29.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f066ef22bc156b3ab525b8202a526c310777bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f066ef22bc156b3ab525b8202a526c310777bad"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c3b7c58a9ed05f98e290ec26019fb46f277ebd68","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1af8a363321f4d62f43cc25d1abf7c7328a74c3c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2015-03-05 10:47:17.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/13"}}},"commit":{"parents":[{"commit":"09ab87151b8fb705a4811373450f66010e10ed8b","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09ab87151b8fb705a4811373450f66010e10ed8b"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-05 10:36:23.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1af8a363321f4d62f43cc25d1abf7c7328a74c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1af8a363321f4d62f43cc25d1abf7c7328a74c3c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"09ab87151b8fb705a4811373450f66010e10ed8b","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"820135eb1274be58e0b10e8070e543b9c0ec3188":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2015-03-05 19:14:58.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/14"}}},"commit":{"parents":[{"commit":"eee3871f47eb1d3595c0cc94a80c24ba3cd86bfc","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eee3871f47eb1d3595c0cc94a80c24ba3cd86bfc"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-05 19:12:51.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/820135eb1274be58e0b10e8070e543b9c0ec3188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/820135eb1274be58e0b10e8070e543b9c0ec3188"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eee3871f47eb1d3595c0cc94a80c24ba3cd86bfc","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ff513be87f83e7ca40cab5acc5ce5ee529618f64":{"kind":"REWORK","_number":15,"created":"2015-03-06 13:01:34.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/15"}}},"commit":{"parents":[{"commit":"71f2eb688d103d8b4c9545aab3dccb1983f78abf","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/71f2eb688d103d8b4c9545aab3dccb1983f78abf"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-06 12:51:15.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ff513be87f83e7ca40cab5acc5ce5ee529618f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ff513be87f83e7ca40cab5acc5ce5ee529618f64"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"71f2eb688d103d8b4c9545aab3dccb1983f78abf","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"309c0d4a83593f97f83528141b35b5d67691ca62":{"kind":"REWORK","_number":16,"created":"2015-03-13 11:48:37.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/16"}}},"commit":{"parents":[{"commit":"7a6d9231f13332f8fcc96c17a49edbe0e456d44a","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a6d9231f13332f8fcc96c17a49edbe0e456d44a"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-13 11:47:19.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/309c0d4a83593f97f83528141b35b5d67691ca62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/309c0d4a83593f97f83528141b35b5d67691ca62"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7a6d9231f13332f8fcc96c17a49edbe0e456d44a","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8f86527946674940c126242acce172139a970de5":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2015-03-13 19:40:47.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/17"}}},"commit":{"parents":[{"commit":"eb82bfcb61075f3d30b27b39c74a48c035b50f24","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb82bfcb61075f3d30b27b39c74a48c035b50f24"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-13 19:27:11.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f86527946674940c126242acce172139a970de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f86527946674940c126242acce172139a970de5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb82bfcb61075f3d30b27b39c74a48c035b50f24","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c3ca63e27d3350324e55a14eeb29da5e4cef066e":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2015-03-18 19:58:24.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/18"}}},"commit":{"parents":[{"commit":"1dcbed1ac70c66dfb3c2cbb6f069becbe55c92d0","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dcbed1ac70c66dfb3c2cbb6f069becbe55c92d0"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-18 19:31:39.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3ca63e27d3350324e55a14eeb29da5e4cef066e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3ca63e27d3350324e55a14eeb29da5e4cef066e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1dcbed1ac70c66dfb3c2cbb6f069becbe55c92d0","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d91916058a63e4c716d8fbaebad2b2b45d718701":{"kind":"REWORK","_number":19,"created":"2015-03-20 08:32:09.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/19"}}},"commit":{"parents":[{"commit":"493e53241c7d2b776be36ddbdc8170110596971f","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/493e53241c7d2b776be36ddbdc8170110596971f"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-20 08:28:29.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d91916058a63e4c716d8fbaebad2b2b45d718701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d91916058a63e4c716d8fbaebad2b2b45d718701"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"493e53241c7d2b776be36ddbdc8170110596971f","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":18,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dde4e012349a7e48131c825e12a33b634b2620bf":{"kind":"REWORK","_number":20,"created":"2015-03-23 17:58:12.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/20"}}},"commit":{"parents":[{"commit":"b9d4e10deb11a74f0b1392ec2836275be99b924d","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b9d4e10deb11a74f0b1392ec2836275be99b924d"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-03-23 17:50:01.000000000","tz":60},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nRelated to blueprint kilo-objects\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dde4e012349a7e48131c825e12a33b634b2620bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dde4e012349a7e48131c825e12a33b634b2620bf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b9d4e10deb11a74f0b1392ec2836275be99b924d","is_merged_in_target_branch":false,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":19,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6285d6a5114b739697f8d0041e31b0b945facf7c":{"kind":"REWORK","_number":21,"created":"2015-04-13 17:49:42.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/64/133664/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/133664/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/133664/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/133664/21"}}},"commit":{"parents":[{"commit":"19455514d1366f60c37ebd5e27da675cc7459f80","subject":"Convert service and servicegroup to objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/19455514d1366f60c37ebd5e27da675cc7459f80"}]}],"author":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2014-11-07 13:52:23.000000000","tz":60},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-04-13 17:47:32.000000000","tz":120},"subject":"Remove and deprecate unused conductor service calls","message":"Remove and deprecate unused conductor service calls\n\nA previous change converted service and servicegroup to objects.\nThis change removes the now unused service methods from conductor.\n\nChange-Id: I3e35598ec89b18d1121547caa37ec5dbc506cb06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6285d6a5114b739697f8d0041e31b0b945facf7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6285d6a5114b739697f8d0041e31b0b945facf7c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19455514d1366f60c37ebd5e27da675cc7459f80","is_merged_in_target_branch":true,"change_id":"I87b077699c2ecbc25a83e5844e8c69fe147cec01","change_number":133663,"patch_set_number":20,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
