)]}'
{"id":"openstack%2Fnova~153426","triplet_id":"openstack%2Fnova~master~Id2c3e7877fd7a9765d3dc0ed5892e35658895468","project":"openstack/nova","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"Id2c3e7877fd7a9765d3dc0ed5892e35658895468","subject":"Updated from global requirements","status":"MERGED","created":"2015-02-06 00:05:20.000000000","updated":"2015-02-06 21:54:54.000000000","submitted":"2015-02-06 21:54:51.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9c509f548021bf2afedd04ce4eddeb280269290","_number":153426,"virtual_id_number":153426,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2015-02-06 20:05:59.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-02-06 21:37:50.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-02-06 21:54:51.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-02-06 19:35:12.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2015-02-06 19:28:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"date":"2015-02-06 20:19:01.000000000","_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2015-02-06 19:15:08.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":2,"date":"2015-02-06 19:04:18.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2015-02-06 19:15:08.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-02-06 12:35:21.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2015-02-06 19:15:08.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2015-02-06 19:28:09.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-06 19:35:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-06 20:05:59.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-06 20:19:01.000000000","updated_by":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-06 21:37:50.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-06 21:54:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5e3748bc7f724622d81fd422364e4c30004fdf15","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2015-02-06 00:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ac268a6149623bb5501e156779ab55882bacec3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-06 00:31:07.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5f7ea8e9d96f4e39bf51b1e64e49b2aa/index.html : SUCCESS in 19m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f972ff553b644301a8d1c6d41699ecdc/index.html : SUCCESS in 22m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a542d418353e4711b2ec2a757af9f6d8/index.html : SUCCESS in 20m 23s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/1/check/real-db-upgrade_nova_percona_user_001:th-percona/5b47825d976e4c37997ca213d26488c5/index.html : SUCCESS in 24m 15s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3773a17b3ad24500a18c40c859fdb6d4/index.html : SUCCESS in 20m 10s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/1/check/real-db-upgrade_nova_percona_user_002:th-percona/da4d4f5b06544d0dbf79f8e235130385/index.html : SUCCESS in 20m 34s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e9c16d14712cfcd2f90db78e74ad473e561715d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-02-06 00:44:36.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/153426/1/check-ibm-tempest-dsvm-full/89b29d1 : SUCCESS in 29m 26s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/153426/1/check-ibm-tempest-dsvm-postgres-full/c7d1c85 : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a4ea38dd19cdff5d240612342e5f36cd5d30b58","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-02-06 00:57:33.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/153426/1/ : Test run failed. in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ea97fa3264285f4c13a67525f08a54465e29c34c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-06 00:59:28.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/153426/1/8139/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"512120d5c4f81a88b0d73521ad2c1c751d751c4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 01:01:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/26/153426/1/check/gate-nova-pep8/a1e9c72/ : SUCCESS in 7m 50s\n- gate-nova-docs http://docs-draft.openstack.org/26/153426/1/check/gate-nova-docs/8ca3a96//doc/build/html/ : SUCCESS in 8m 06s\n- gate-nova-python27 http://logs.openstack.org/26/153426/1/check/gate-nova-python27/179131e/ : SUCCESS in 12m 27s\n- gate-nova-requirements http://logs.openstack.org/26/153426/1/check/gate-nova-requirements/0632b13/ : SUCCESS in 51s\n- check-tempest-dsvm-full http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-full/374f56a/ : SUCCESS in 42m 09s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-postgres-full/58259df/ : SUCCESS in 35m 58s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-neutron-full/00d99a2/ : SUCCESS in 55m 02s\n- check-grenade-dsvm http://logs.openstack.org/26/153426/1/check/check-grenade-dsvm/b16aa45/ : SUCCESS in 38m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/26/153426/1/check/gate-tempest-dsvm-large-ops/607a621/ : SUCCESS in 23m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/26/153426/1/check/gate-tempest-dsvm-neutron-large-ops/0802221/ : SUCCESS in 19m 39s\n- check-devstack-dsvm-cells http://logs.openstack.org/26/153426/1/check/check-devstack-dsvm-cells/09ca717/ : SUCCESS in 17m 44s\n- gate-nova-tox-functional http://logs.openstack.org/26/153426/1/check/gate-nova-tox-functional/0ceb5ed/ : SUCCESS in 11m 36s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/26/153426/1/check/check-grenade-dsvm-partial-ncpu/e5df5f7/ : SUCCESS in 34m 24s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-ironic-pxe_ssh/11aeaa1/ : SUCCESS in 34m 24s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-nova-v21-full/196d433/ : SUCCESS in 45m 03s\n- check-tempest-dsvm-cells http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-cells/8582eb5/ : FAILURE in 33m 56s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/26/153426/1/check/check-tempest-dsvm-full-ceph/55086f0/ : SUCCESS in 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"917c8641a875bf3b814abbf9fd05deb958913821","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-02-06 02:45:21.000000000","message":"Patch Set 1:\n\nBuild Successful. Logs at http://198.175.100.33/153426/1","accounts_in_message":[],"_revision_number":1},{"id":"fcddfc6a6fe0ee059dfbc5a8844a681655fce6f6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2015-02-06 12:21:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1724e4d93e977068779cabdf642170618a947ad3","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-02-06 12:35:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ec8596968d8bdd3f0ea6325d66a0d80cb996101","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-06 12:44:49.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5caf71532dd34099a051fb89a715fa17/index.html : SUCCESS in 16m 26s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a0e44cd5f2fc400390788ed5e4795cfe/index.html : SUCCESS in 21m 52s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1eb58953ccbb4c7f81497d31728490f2/index.html : SUCCESS in 19m 43s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/2/check/real-db-upgrade_nova_percona_user_001:th-percona/8bd7b0c38aa045e6b663e83f77706e7b/index.html : SUCCESS in 19m 25s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/43c325351b8a4f49a984fa304cb78528/index.html : SUCCESS in 17m 18s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/2/check/real-db-upgrade_nova_percona_user_002:th-percona/3c70e9be8b1c4a1885d7a6f2e051d3dd/index.html : SUCCESS in 17m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"879c8b15ea97a5737cf1d97c925b33097712d562","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-02-06 13:09:36.000000000","message":"Patch Set 2:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/153426/2/ : Test run failed. in 47m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a00eeb2216da96fe226df4aecc620cf3bb3a93e2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 13:13:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/26/153426/2/check/gate-nova-pep8/e01ff38/ : SUCCESS in 7m 47s\n- gate-nova-docs http://docs-draft.openstack.org/26/153426/2/check/gate-nova-docs/688f299//doc/build/html/ : SUCCESS in 10m 30s\n- gate-nova-python27 http://logs.openstack.org/26/153426/2/check/gate-nova-python27/a8e451d/ : SUCCESS in 12m 08s\n- gate-nova-requirements http://logs.openstack.org/26/153426/2/check/gate-nova-requirements/65b303a/ : SUCCESS in 1m 18s\n- check-tempest-dsvm-full http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-full/67ec85e/ : SUCCESS in 40m 32s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-postgres-full/ca7754f/ : SUCCESS in 50m 36s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-neutron-full/fcd230c/ : SUCCESS in 49m 30s\n- check-grenade-dsvm http://logs.openstack.org/26/153426/2/check/check-grenade-dsvm/fcb98bb/ : SUCCESS in 33m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/26/153426/2/check/gate-tempest-dsvm-large-ops/9bf682f/ : SUCCESS in 19m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/26/153426/2/check/gate-tempest-dsvm-neutron-large-ops/38009ee/ : SUCCESS in 20m 48s\n- check-devstack-dsvm-cells http://logs.openstack.org/26/153426/2/check/check-devstack-dsvm-cells/6c1d3bc/ : SUCCESS in 15m 47s\n- gate-nova-tox-functional http://logs.openstack.org/26/153426/2/check/gate-nova-tox-functional/ce682ec/ : SUCCESS in 10m 51s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/26/153426/2/check/check-grenade-dsvm-partial-ncpu/5006682/ : SUCCESS in 32m 00s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-ironic-pxe_ssh/7c33dbe/ : SUCCESS in 28m 22s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-nova-v21-full/c1d4d1a/ : SUCCESS in 46m 02s\n- check-tempest-dsvm-cells http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-cells/03b205f/ : FAILURE in 29m 57s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/26/153426/2/check/check-tempest-dsvm-full-ceph/55651f6/ : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"450b09e360e128d60e938622ecd8fc17be3b7009","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 13:13:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"626b56f4d8af744c15dc7164487652ac334a45e8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-06 13:30:33.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/153426/2/8211/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"28fdcdb3124ba529f07404380b1b87f55d7a883a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-02-06 13:49:52.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/153426/2/check-ibm-tempest-dsvm-full/fea5409 : SUCCESS in 40m 06s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/153426/2/check-ibm-tempest-dsvm-postgres-full/5505d50 : SUCCESS in 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb9c162ed83ff783b6eed2549296df141015ffaf","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-02-06 13:56:34.000000000","message":"Patch Set 2:\n\nBuild Successful. Logs at http://198.175.100.33/153426/2","accounts_in_message":[],"_revision_number":2},{"id":"52c72e5136f4cc18b84bc007e95f77f4abb6eb5f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 16:22:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/153426/2/gate/gate-nova-docs/e8835f7//doc/build/html/ : SUCCESS in 10m 11s\n- gate-nova-pep8 http://logs.openstack.org/26/153426/2/gate/gate-nova-pep8/5abcddf/ : SUCCESS in 10m 43s\n- gate-nova-python27 http://logs.openstack.org/26/153426/2/gate/gate-nova-python27/00e3497/ : SUCCESS in 16m 08s\n- gate-nova-requirements http://logs.openstack.org/26/153426/2/gate/gate-nova-requirements/3fa34b6/ : SUCCESS in 53s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/153426/2/gate/gate-tempest-dsvm-full/e4ad6df/ : SUCCESS in 36m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/153426/2/gate/gate-tempest-dsvm-postgres-full/83261b1/ : SUCCESS in 38m 24s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/153426/2/gate/gate-tempest-dsvm-neutron-full/f71a7cb/ : SUCCESS in 55m 39s\n- gate-grenade-dsvm http://logs.openstack.org/26/153426/2/gate/gate-grenade-dsvm/a031df0/ : SUCCESS in 32m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/26/153426/2/gate/gate-tempest-dsvm-large-ops/d9e0a3c/ : SUCCESS in 20m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/26/153426/2/gate/gate-tempest-dsvm-neutron-large-ops/73a085b/ : SUCCESS in 18m 11s\n- gate-nova-tox-functional http://logs.openstack.org/26/153426/2/gate/gate-nova-tox-functional/c1c52a2/ : SUCCESS in 11m 18s\n- gate-devstack-dsvm-cells http://logs.openstack.org/26/153426/2/gate/gate-devstack-dsvm-cells/b6ab9f5/ : SUCCESS in 16m 40s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/26/153426/2/gate/gate-grenade-dsvm-partial-ncpu/b0e4565/ : SUCCESS in 29m 02s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/26/153426/2/gate/gate-tempest-dsvm-nova-v21-full/c80cf2c/ : FAILURE in 39m 35s","accounts_in_message":[],"_revision_number":2},{"id":"325367a989e5001a32db9cc748ae750cedfae3bd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2015-02-06 19:04:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"42e1b68c77baf14b9db472e5d25504ad20b30b68","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-02-06 19:15:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a9f3a0bad1f44a86e86a264e7a5b5189bc379cbc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-06 19:28:09.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a3820749c343416d9501271960c79a81/index.html : SUCCESS in 16m 05s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6e9a45e682814776826b75054d28c2d8/index.html : SUCCESS in 22m 02s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1b3698a6a83e4cfeb474b45deaef4607/index.html : SUCCESS in 19m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/3/check/real-db-upgrade_nova_percona_user_001:th-percona/5823c2f21484477098fe96605eb28f6f/index.html : SUCCESS in 22m 33s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/153426/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2da757f9b08940e5a8af2d1fd7aaed15/index.html : SUCCESS in 16m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/153426/3/check/real-db-upgrade_nova_percona_user_002:th-percona/59d6e018d1454d378299d9be191684eb/index.html : SUCCESS in 19m 46s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"1deb469679cbd6860d611ab065422a50f36bf762","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-02-06 19:35:12.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/153426/3/check-ibm-tempest-dsvm-full/897465a : SUCCESS in 29m 18s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/153426/3/check-ibm-tempest-dsvm-postgres-full/60abeb1 : SUCCESS in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f984f1d59805e52743de76908c6a114c6858292","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-06 20:05:59.000000000","message":"Patch Set 3: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/153426/3/8253/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"6e877b565fb901edbeb4e0497269e7e8ef3ef5f7","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-02-06 20:19:01.000000000","message":"Patch Set 3:\n\nBuild Successful. Logs at http://198.175.100.33/153426/3","accounts_in_message":[],"_revision_number":3},{"id":"aa00d5f01e6b9f7d69ba356b6c451f62433facd2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 20:44:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/26/153426/3/check/gate-nova-pep8/0116a56/ : SUCCESS in 10m 33s\n- gate-nova-docs http://docs-draft.openstack.org/26/153426/3/check/gate-nova-docs/2ec1f51//doc/build/html/ : SUCCESS in 10m 33s\n- gate-nova-python27 http://logs.openstack.org/26/153426/3/check/gate-nova-python27/86248dc/ : SUCCESS in 15m 10s\n- gate-nova-requirements http://logs.openstack.org/26/153426/3/check/gate-nova-requirements/36fe2ad/ : SUCCESS in 53s\n- check-tempest-dsvm-full http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-full/9d4cd5e/ : SUCCESS in 45m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-postgres-full/4d50720/ : SUCCESS in 45m 11s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-neutron-full/399cb89/ : SUCCESS in 49m 44s\n- check-grenade-dsvm http://logs.openstack.org/26/153426/3/check/check-grenade-dsvm/863f03b/ : SUCCESS in 34m 58s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/26/153426/3/check/gate-tempest-dsvm-large-ops/7d329e7/ : SUCCESS in 23m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/26/153426/3/check/gate-tempest-dsvm-neutron-large-ops/6309856/ : SUCCESS in 17m 30s\n- check-devstack-dsvm-cells http://logs.openstack.org/26/153426/3/check/check-devstack-dsvm-cells/be75d48/ : SUCCESS in 15m 43s\n- gate-nova-tox-functional http://logs.openstack.org/26/153426/3/check/gate-nova-tox-functional/d6b650a/ : SUCCESS in 14m 28s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/26/153426/3/check/check-grenade-dsvm-partial-ncpu/9cf7488/ : SUCCESS in 33m 03s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-ironic-pxe_ssh/5e27fd6/ : SUCCESS in 32m 45s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-nova-v21-full/b763b10/ : SUCCESS in 41m 34s\n- check-tempest-dsvm-cells http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-cells/1222a7d/ : FAILURE in 31m 08s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/26/153426/3/check/check-tempest-dsvm-full-ceph/d03657d/ : SUCCESS in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"98735a40276b6b8f2ed3f97cc3ea1591b136739a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 20:48:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"aa4399ce501cbcec54c6ab673d2f304a0dac28af","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-02-06 21:37:50.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/153426/3/ : Test run failed. in 2h 32m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8cb990b74fbd600314b652c79b3afbda4edc557e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 21:54:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/153426/3/gate/gate-nova-docs/a309953//doc/build/html/ : SUCCESS in 7m 30s\n- gate-nova-pep8 http://logs.openstack.org/26/153426/3/gate/gate-nova-pep8/8ecbb16/ : SUCCESS in 7m 54s\n- gate-nova-python27 http://logs.openstack.org/26/153426/3/gate/gate-nova-python27/b30c796/ : SUCCESS in 11m 47s\n- gate-nova-requirements http://logs.openstack.org/26/153426/3/gate/gate-nova-requirements/eeb5713/ : SUCCESS in 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/153426/3/gate/gate-tempest-dsvm-full/5d3be07/ : SUCCESS in 39m 03s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/153426/3/gate/gate-tempest-dsvm-postgres-full/4524d40/ : SUCCESS in 37m 19s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/153426/3/gate/gate-tempest-dsvm-neutron-full/cbbd182/ : SUCCESS in 51m 27s\n- gate-grenade-dsvm http://logs.openstack.org/26/153426/3/gate/gate-grenade-dsvm/e7c2249/ : SUCCESS in 35m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/26/153426/3/gate/gate-tempest-dsvm-large-ops/1794974/ : SUCCESS in 19m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/26/153426/3/gate/gate-tempest-dsvm-neutron-large-ops/c7f7dfc/ : SUCCESS in 22m 26s\n- gate-nova-tox-functional http://logs.openstack.org/26/153426/3/gate/gate-nova-tox-functional/0cd26bf/ : SUCCESS in 11m 11s\n- gate-devstack-dsvm-cells http://logs.openstack.org/26/153426/3/gate/gate-devstack-dsvm-cells/df05b49/ : SUCCESS in 15m 57s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/26/153426/3/gate/gate-grenade-dsvm-partial-ncpu/8703591/ : SUCCESS in 37m 20s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/26/153426/3/gate/gate-tempest-dsvm-nova-v21-full/9788d4b/ : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":3},{"id":"09d6c197cf98307ba20a91f29c3bd80d81b60175","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-06 21:54:54.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"027579779f56ccf6a8fc0459222e53f3e810bee1","revisions":{"05a3f833912cb067c9aeae685e198c387fe782e0":{"kind":"REWORK","_number":1,"created":"2015-02-06 00:05:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/153426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/153426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/153426/1"}}},"commit":{"parents":[{"commit":"98641da6402c44793da4d2f6d7ea1a7ba5f16708","subject":"Merge \"Ironic: Adds config drive support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/98641da6402c44793da4d2f6d7ea1a7ba5f16708"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-02-06 00:05:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-02-06 00:05:16.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Id2c3e7877fd7a9765d3dc0ed5892e35658895468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05a3f833912cb067c9aeae685e198c387fe782e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05a3f833912cb067c9aeae685e198c387fe782e0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"98641da6402c44793da4d2f6d7ea1a7ba5f16708","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"798e9a960dd2ff155fb2a15cd18c76e5152bd624":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2015-02-06 12:21:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/153426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/153426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/153426/2"}}},"commit":{"parents":[{"commit":"ff219370f3e1c836603b218dd3225f81ba3938b4","subject":"Merge \"Fix references to non-existent \"pause\" section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ff219370f3e1c836603b218dd3225f81ba3938b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-02-06 12:21:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-02-06 12:21:44.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Id2c3e7877fd7a9765d3dc0ed5892e35658895468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/798e9a960dd2ff155fb2a15cd18c76e5152bd624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/798e9a960dd2ff155fb2a15cd18c76e5152bd624"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ff219370f3e1c836603b218dd3225f81ba3938b4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"027579779f56ccf6a8fc0459222e53f3e810bee1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2015-02-06 19:04:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/153426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/153426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/153426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/153426/3"}}},"commit":{"parents":[{"commit":"cb3d079343d1bd1d0adbf0fb3d5350c00d42a1b7","subject":"Merge \"Fix conductor servicegroup joining when zk driver is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb3d079343d1bd1d0adbf0fb3d5350c00d42a1b7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-02-06 19:04:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2015-02-06 19:04:14.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: Id2c3e7877fd7a9765d3dc0ed5892e35658895468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/027579779f56ccf6a8fc0459222e53f3e810bee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/027579779f56ccf6a8fc0459222e53f3e810bee1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cb3d079343d1bd1d0adbf0fb3d5350c00d42a1b7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
