)]}'
{"id":"openstack%2Fnova~157731","triplet_id":"openstack%2Fnova~master~I50480121fd906b929571759f537426a8124ed1bd","project":"openstack/nova","branch":"master","topic":"osapi_compute_unique_server_name_scope","hashtags":[],"change_id":"I50480121fd906b929571759f537426a8124ed1bd","subject":"Deprecate osapi_compute_unique_server_name_scope","status":"ABANDONED","created":"2015-02-20 11:54:46.000000000","updated":"2015-02-23 09:43:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c8b7ac16a612858de3e3e5c839db1ffcff851cf3","_number":157731,"virtual_id_number":157731,"owner":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"all":[{"date":"2015-02-22 07:36:02.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2015-02-20 12:55:54.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-02-20 13:11:48.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-02-20 12:25:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-02-20 12:25:25.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-20 12:55:54.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-02-20 13:11:48.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-02-22 07:36:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6f851e658919cc78d027e552cdba5bd6d5bd82d","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2015-02-20 11:54:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4ad72c55b2b054b0e3aa5aaf3487d7518de5eb4","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-02-20 12:25:25.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/31/157731/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2611d29c98db4af4b963598263fe37a3/index.html : SUCCESS in 20m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/31/157731/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/355ae721d7d84b1b995cd3b3d3bf8844/index.html : SUCCESS in 23m 24s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/31/157731/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6a17b588958d4c77a1486dbe0fc81d9b/index.html : SUCCESS in 20m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/31/157731/1/check/real-db-upgrade_nova_percona_user_001:th-percona/443a60a114704e04aba8d3eb639e5be0/index.html : SUCCESS in 23m 35s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/31/157731/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1e210bb6346043558713ef6e40f96379/index.html : SUCCESS in 20m 42s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/31/157731/1/check/real-db-upgrade_nova_percona_user_002:th-percona/576c1772ca7c453f86b320ebef6eca30/index.html : SUCCESS in 21m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"4a4881e9f8ee509d21ede5adb7d08c3282708861","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-02-20 12:55:54.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/31/157731/1/10020/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"e91254a58cffb5ce509445ec42d516906938e8b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-02-20 13:11:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/31/157731/1/check/gate-nova-pep8/d073bd4/ : SUCCESS in 8m 38s\n- gate-nova-docs http://docs-draft.openstack.org/31/157731/1/check/gate-nova-docs/36eeb09//doc/build/html/ : SUCCESS in 7m 48s\n- gate-nova-python27 http://logs.openstack.org/31/157731/1/check/gate-nova-python27/8be8d19/ : SUCCESS in 12m 04s\n- check-tempest-dsvm-full http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-full/ffb23ba/ : SUCCESS in 58m 01s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-postgres-full/9db5f5b/ : SUCCESS in 54m 42s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-neutron-full/8ae6666/ : SUCCESS in 1h 11m 34s\n- check-grenade-dsvm http://logs.openstack.org/31/157731/1/check/check-grenade-dsvm/7e5993d/ : SUCCESS in 42m 27s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/31/157731/1/check/gate-tempest-dsvm-large-ops/269335e/ : SUCCESS in 27m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/31/157731/1/check/gate-tempest-dsvm-neutron-large-ops/2242cb3/ : SUCCESS in 26m 34s\n- check-devstack-dsvm-cells http://logs.openstack.org/31/157731/1/check/check-devstack-dsvm-cells/6bca9cc/ : SUCCESS in 22m 42s\n- gate-nova-tox-functional http://logs.openstack.org/31/157731/1/check/gate-nova-tox-functional/212631c/ : SUCCESS in 10m 43s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/31/157731/1/check/check-grenade-dsvm-partial-ncpu/792ec24/ : SUCCESS in 40m 48s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-ironic-pxe_ssh/994bdd4/ : SUCCESS in 37m 59s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-nova-v21-full/0e91cac/ : SUCCESS in 52m 40s\n- check-tempest-dsvm-cells http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-cells/3be3e1c/ : FAILURE in 37m 45s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/31/157731/1/check/check-tempest-dsvm-full-ceph/ccb7e56/ : SUCCESS in 49m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef1c65cb18101e8e69f913f2832aecefd090450b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-02-22 07:36:02.000000000","message":"Patch Set 1:\n\nBuild failed. Logs: http://208.91.1.172/logs/157731/1 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":1},{"id":"0f991be89029c70488f42f1178c034db7a737ea7","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2015-02-23 09:43:21.000000000","message":"Abandoned\n\nIdentified 2 current users of this feature, so deprecation with no replacement isn\u0027t realistic.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0bbbdf7f76da6760b70a9162c21c363bf91349a","revisions":{"f0bbbdf7f76da6760b70a9162c21c363bf91349a":{"kind":"REWORK","_number":1,"created":"2015-02-20 11:54:46.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/31/157731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/157731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/157731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/157731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/157731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/157731/1"}}},"commit":{"parents":[{"commit":"c4a9cc56b2e783f5968c2696377ba8d488ed357d","subject":"Merge \"Fix VNC access, when reverse DNS lookups fail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4a9cc56b2e783f5968c2696377ba8d488ed357d"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2015-02-19 15:27:32.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2015-02-19 15:30:42.000000000","tz":0},"subject":"Deprecate osapi_compute_unique_server_name_scope","message":"Deprecate osapi_compute_unique_server_name_scope\n\nThis behaviour is difficult to support safely, and does not appear to\nbe widely used.\n\nChange-Id: I50480121fd906b929571759f537426a8124ed1bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f0bbbdf7f76da6760b70a9162c21c363bf91349a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f0bbbdf7f76da6760b70a9162c21c363bf91349a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c4a9cc56b2e783f5968c2696377ba8d488ed357d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
