)]}'
{"id":"openstack%2Fnova~162470","triplet_id":"openstack%2Fnova~master~I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7","project":"openstack/nova","branch":"master","topic":"bp/vif-plugin-script","hashtags":[],"change_id":"I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7","subject":"WIP: Add support for VIF plugin script to Nova","status":"ABANDONED","created":"2015-03-08 13:21:45.000000000","updated":"2015-06-25 14:49:46.000000000","total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"29044a7bbb148c3efe58b95f81bbe4fe4b66a56d","_number":162470,"virtual_id_number":162470,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2015-05-27 17:34:54.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2015-05-27 18:04:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"date":"2015-05-27 17:35:09.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"date":"2015-05-27 16:48:01.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2015-05-28 01:26:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"date":"2015-06-02 16:19:31.000000000","_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},{"value":1,"date":"2015-05-28 15:36:17.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2015-05-27 17:38:56.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"date":"2015-06-04 17:45:28.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},{"value":1,"date":"2015-05-28 03:29:55.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2015-05-27 17:36:54.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-05-27 17:35:13.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-03-08 22:02:26.000000000","updated_by":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 18:54:15.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-08 21:52:06.000000000","updated_by":{"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},"reviewer":{"_account_id":12340,"name":"Vincent Jardin","email":"vincent.jardin@6wind.com","username":"vjardin"},"state":"REVIEWER"},{"updated":"2015-05-12 17:22:43.000000000","updated_by":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"reviewer":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"state":"REVIEWER"},{"updated":"2015-05-20 12:59:50.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2015-05-22 16:53:48.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2015-05-26 20:03:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-27 16:48:01.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-27 17:34:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-05-27 17:35:09.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-05-27 17:36:54.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2015-05-27 17:38:56.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-27 18:04:47.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-28 01:26:01.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-28 03:29:55.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-05-28 15:36:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-06-02 16:19:31.000000000","updated_by":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"reviewer":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"state":"REVIEWER"},{"updated":"2015-06-04 17:45:28.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"d315ab7a397526e3af6812372f3b83c3814b401e","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-08 13:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14c6ba9be7d574a653d35210f63a9b5f1c04c426","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-08 13:22:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2d6e6932712369ccff341438468f9ada77ec111b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-08 13:46:33.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/93f55eee50ec46f3b7041649c022af45/index.html : SUCCESS in 20m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/91b0d4e47a054da397334e364b993cf5/index.html : SUCCESS in 23m 21s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ef8fd8f6278a40b79e983f96bdb54b61/index.html : SUCCESS in 20m 18s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/1/check/real-db-upgrade_nova_percona_user_001:th-percona/d5c5470984924eb48fd81802df7b9f00/index.html : SUCCESS in 23m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/444abc044be24e6ca61228ad09e73936/index.html : SUCCESS in 19m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/1/check/real-db-upgrade_nova_percona_user_002:th-percona/341ad5d3ab474c00be3893520c7dccf5/index.html : SUCCESS in 17m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"3755df5690e9f53b3acc4cc342ade45fc63ba9c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-08 13:54:12.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/1/check-ibm-tempest-dsvm-full/43ddc51 : SUCCESS in 29m 15s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/1/check-ibm-tempest-dsvm-postgres-full/c785d20 : SUCCESS in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8757a09b6b19802413607157d440c2e3bf8cced7","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-08 14:00:22.000000000","message":"Patch Set 1:\n\nBuild Successful. Logs at http://198.175.100.33/162470/1","accounts_in_message":[],"_revision_number":1},{"id":"42adb94f967340b82d64b6bbaf09827e9f0986e9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-08 14:09:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/70/162470/1/check/gate-nova-pep8/5cc44f5/ : SUCCESS in 7m 27s\n- gate-nova-docs http://docs-draft.openstack.org/70/162470/1/check/gate-nova-docs/acca8ab//doc/build/html/ : SUCCESS in 7m 27s\n- gate-nova-python27 http://logs.openstack.org/70/162470/1/check/gate-nova-python27/ae3f8f8/ : SUCCESS in 15m 32s\n- check-tempest-dsvm-full http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-full/f63526e/ : SUCCESS in 46m 08s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-postgres-full/af0f819/ : SUCCESS in 45m 03s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-neutron-full/0ee7aa5/ : SUCCESS in 46m 41s\n- check-grenade-dsvm http://logs.openstack.org/70/162470/1/check/check-grenade-dsvm/dfd45f4/ : SUCCESS in 34m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/70/162470/1/check/gate-tempest-dsvm-large-ops/f4b27ac/ : SUCCESS in 23m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/70/162470/1/check/gate-tempest-dsvm-neutron-large-ops/5dcfcce/ : SUCCESS in 21m 52s\n- check-devstack-dsvm-cells http://logs.openstack.org/70/162470/1/check/check-devstack-dsvm-cells/bf8eae3/ : SUCCESS in 18m 00s\n- gate-nova-tox-functional http://logs.openstack.org/70/162470/1/check/gate-nova-tox-functional/e57b0d3/ : SUCCESS in 11m 00s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/70/162470/1/check/check-grenade-dsvm-partial-ncpu/2d9c88d/ : SUCCESS in 37m 50s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-ironic-pxe_ssh/0dacefc/ : SUCCESS in 32m 56s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-nova-v21-full/fd0a96b/ : SUCCESS in 40m 15s\n- check-tempest-dsvm-cells http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-cells/52cb140/ : FAILURE in 36m 19s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/70/162470/1/check/check-tempest-dsvm-full-ceph/324799a/ : SUCCESS in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39d68f79a6f998e83e3095633bb4655040f0348b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-08 14:13:59.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/70/162470/1/12221/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"a6810b489d8f689f27393ec71bcd1c5fe6c4607e","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-08 21:48:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3e974ce2f13164a15f70d2690cb4911f1c9ae8a","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-08 21:48:59.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"31f0aeee6c4a9c95c62fc8f8651a471973a042b6","author":{"_account_id":11103,"name":"Intel PCI CI","email":"otc_cloud@163.com","username":"intelotccloud","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-03-08 22:02:26.000000000","message":"Patch Set 2:\n\nBuild Successful. Logs at http://198.175.100.33/162470/2","accounts_in_message":[],"_revision_number":2},{"id":"81f498ad12ffe908b5b84644f72a625a966c173d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-08 22:14:16.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/cf2f038bde104237ae5d9928c31f9fc3/index.html : SUCCESS in 20m 37s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d7dc5c106c634c0c9b273f78670bd326/index.html : SUCCESS in 20m 02s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/bfdda1b94c5a48aab12f9a8ff8c99c90/index.html : SUCCESS in 20m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/2/check/real-db-upgrade_nova_percona_user_001:th-percona/006843bcc1b54384ade338b3aa90eccd/index.html : SUCCESS in 20m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/bc7119a3513e488f8b6f87c881a45ac1/index.html : SUCCESS in 24m 14s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/2/check/real-db-upgrade_nova_percona_user_002:th-percona/e4b2bd0d72ad499498acb835df7e1261/index.html : SUCCESS in 20m 56s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ab90ab3e3559bbf63340e165d29a6c0fb3fce49","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-08 22:25:24.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/2/check-ibm-tempest-dsvm-full/3bdfd87 : SUCCESS in 36m 11s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/2/check-ibm-tempest-dsvm-postgres-full/85edb3b : FAILURE in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f640f5e99f94eaa7f978e10cd5abb5a691a6741","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-08 22:40:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/70/162470/2/check/gate-nova-pep8/ec698a7/ : SUCCESS in 7m 57s\n- gate-nova-docs http://docs-draft.openstack.org/70/162470/2/check/gate-nova-docs/26ba31b//doc/build/html/ : SUCCESS in 8m 23s\n- gate-nova-python27 http://logs.openstack.org/70/162470/2/check/gate-nova-python27/ac573dd/ : SUCCESS in 11m 54s\n- check-tempest-dsvm-full http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-full/426f5fa/ : SUCCESS in 38m 18s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-postgres-full/59952c7/ : SUCCESS in 41m 22s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-neutron-full/28edd37/ : SUCCESS in 50m 36s\n- check-grenade-dsvm http://logs.openstack.org/70/162470/2/check/check-grenade-dsvm/cddd4cc/ : SUCCESS in 42m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/70/162470/2/check/gate-tempest-dsvm-large-ops/ba21e2b/ : SUCCESS in 20m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/70/162470/2/check/gate-tempest-dsvm-neutron-large-ops/047d9e1/ : SUCCESS in 20m 14s\n- check-devstack-dsvm-cells http://logs.openstack.org/70/162470/2/check/check-devstack-dsvm-cells/9fceddd/ : SUCCESS in 20m 49s\n- gate-nova-tox-functional http://logs.openstack.org/70/162470/2/check/gate-nova-tox-functional/9d4f144/ : SUCCESS in 11m 20s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/70/162470/2/check/check-grenade-dsvm-partial-ncpu/3e72cf5/ : SUCCESS in 43m 58s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-ironic-pxe_ssh/b39c4d3/ : SUCCESS in 39m 37s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-nova-v21-full/8efa3db/ : SUCCESS in 35m 11s\n- check-tempest-dsvm-cells http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-cells/e84931d/ : FAILURE in 31m 22s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/70/162470/2/check/check-tempest-dsvm-full-ceph/8a58f04/ : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6491ef7e874dc74fae9f42a4ad116d8bb47e3d0d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-08 22:48:19.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/70/162470/2/12228/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"8b26bffc5118aa4ab387bc0dcba70b713df8f55a","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-03-09 13:23:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a39731f89cb7fa343d2e2bcf1d5157f66ea3698","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2015-03-09 15:33:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1f4ca86bf7e247beedf9e146ad8606ea3e57fc66","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-09 15:48:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"afe737d2dc074a3ddbef09ede289126c580712f6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-10 00:11:41.000000000","message":"Patch Set 2:\n\nBuild failed. Logs: http://208.91.1.172/logs/162470/2 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":2},{"id":"bfda7ae5f0015b57d7ae4550e7eea53865fa7792","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2015-03-17 18:07:52.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"32789f6e839a36041d9b1debae8f96d3bf91273b","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-25 19:49:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e5e6a9f2930c6deff0c0f0613d7740cdc0462175","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-03-25 19:55:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a2567718d4e5c21395e6297f2855ab67c5d8cc04","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-25 20:21:39.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8ef6c33f2e5b4c7f82ce7f9aa3c18dd8/index.html : SUCCESS in 18m 38s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2ce86f2f3cf548708e2fda9d47da5c09/index.html : SUCCESS in 25m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4682adc3ee3245d3a50c6c596bfa39bb/index.html : SUCCESS in 21m 20s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/4/check/real-db-upgrade_nova_percona_user_001:th-percona/1b7b89ed2b2744f48abf7b2751d9ed5e/index.html : SUCCESS in 20m 56s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/73ef209b12c24c51a44d182cdd48725f/index.html : SUCCESS in 18m 57s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/4/check/real-db-upgrade_nova_percona_user_002:th-percona/99d6f2c087a3478cb465de3bda4f7679/index.html : SUCCESS in 21m 40s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"fcd8623d9d8affd403116f2bf3b89aec15852d81","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-25 20:50:56.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/70/162470/4/14436/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"1f68d02c67951739ec3447a0520cae39abd0a8b8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-25 21:04:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/70/162470/4/check/gate-nova-pep8/858ab72/ : SUCCESS in 8m 36s\n- gate-nova-docs http://docs-draft.openstack.org/70/162470/4/check/gate-nova-docs/6ed5ee3//doc/build/html/ : SUCCESS in 7m 58s\n- gate-nova-python27 http://logs.openstack.org/70/162470/4/check/gate-nova-python27/a543b21/ : SUCCESS in 15m 58s\n- check-tempest-dsvm-full http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-full/99f474b/ : SUCCESS in 37m 02s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-postgres-full/6af411a/ : SUCCESS in 38m 46s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-neutron-full/626d669/ : SUCCESS in 46m 12s\n- check-grenade-dsvm http://logs.openstack.org/70/162470/4/check/check-grenade-dsvm/1a7b610/ : SUCCESS in 41m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/70/162470/4/check/gate-tempest-dsvm-large-ops/03944b1/ : SUCCESS in 27m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/70/162470/4/check/gate-tempest-dsvm-neutron-large-ops/a316fcb/ : SUCCESS in 21m 56s\n- check-devstack-dsvm-cells http://logs.openstack.org/70/162470/4/check/check-devstack-dsvm-cells/4c28b58/ : SUCCESS in 19m 47s\n- gate-nova-tox-functional http://logs.openstack.org/70/162470/4/check/gate-nova-tox-functional/7267b7b/ : SUCCESS in 10m 44s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/70/162470/4/check/check-grenade-dsvm-partial-ncpu/59d68c2/ : SUCCESS in 39m 03s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-ironic-pxe_ssh/161e479/ : SUCCESS in 32m 33s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-nova-v21-full/fd10fab/ : SUCCESS in 41m 23s\n- check-tempest-dsvm-cells http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-cells/7c2a48e/ : FAILURE in 28m 45s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/70/162470/4/check/check-tempest-dsvm-full-ceph/0e170ce/ : SUCCESS in 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3fdef72b1410b963f681ef8c9efba16bc0f1a369","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-25 21:04:58.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/4/check-ibm-tempest-dsvm-full/fe524a2 : SUCCESS in 1h 08m 48s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/4/check-ibm-tempest-dsvm-postgres-full/07792cb : SUCCESS in 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6336aba2cfb64194223871364933a487ba8c490f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-30 19:39:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/162470/4 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/162470/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"a5c6ca4ab73eb644743458a8859b6f4627a1ba28","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-03-30 19:39:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/162470/3 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/162470/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"27a042817897c1f4620ed2e3e1a7ff6cf3dc6106","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2015-04-01 15:47:56.000000000","message":"Patch Set 4:\n\nPatch looks good for me. \n\nI have been able to test it successfully with the following script: https://github.com/6WIND/nova-vif-plug-example/blob/master/vif_ovs_plug_script.py\n\nBut I didn\u0027t need the \"WIP: Refactor Nova VIF model\" patch. \nThus, why this patch depends on \"WIP: Refactor Nova VIF model\" ?","accounts_in_message":[],"_revision_number":4},{"id":"a71f1557fab6301f26a643ba9bd158b643f4535a","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2015-04-17 12:16:01.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nI\u0027ve made a couple of comments, but this basically looks good to me.\n\nI don\u0027t think it really depends on the refactoring of VIFs into classes, does it?","accounts_in_message":[],"_revision_number":4},{"id":"a2f5e461a13b8b1946028a47f55bbb2fda39915b","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-04-17 12:54:06.000000000","message":"Patch Set 4: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cf00a38ca7e3ad283a453e37f618a7b7c4d64977","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-04-21 17:33:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c9e65b18fe1772030f08053110800d615399e66","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-04-21 17:35:17.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"550ed5983f396f4ebbcc2f57e6bc48929e405dac","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-04-21 17:49:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2735634b73180ef212b62552366dfacd5568ea0d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-21 17:50:10.000000000","message":"Patch Set 5:\n\n* pci-test http://198.175.100.33/162470/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c51d0c8e3d87cbe4df9103fcfbc85352e0a6f8c7","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-04-21 17:50:25.000000000","message":"Patch Set 6: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"179a091ed55cdf4a4feded2e0b45938a46f0dc34","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-21 18:05:44.000000000","message":"Patch Set 6:\n\n* pci-test http://198.175.100.33/162470/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"64824dd88edb4751b1093b9ff79370c81c46ce05","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-21 18:09:02.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/70ee790b548f410d8bacfcf458daddcb/index.html : SUCCESS in 16m 39s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7dc6b71d020c4075a3302c67f57ebf25/index.html : SUCCESS in 18m 13s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9019e1f56078499088b0a2c6a9d1531f/index.html : SUCCESS in 13m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/6/check/real-db-upgrade_nova_percona_user_001:th-percona/4efd0f745b564eb987dfc2d361762e9a/index.html : SUCCESS in 16m 34s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6cc4a77f0e67421da585636a212f9f24/index.html : SUCCESS in 14m 30s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/6/check/real-db-upgrade_nova_percona_user_002:th-percona/e834eb4ea1ee4f32894cb3c2183dd8b1/index.html : SUCCESS in 16m 32s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"b060ec8b8b7383984f231d30005379469524b888","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-21 18:43:48.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/6/check-ibm-tempest-dsvm-full/315f984 : SUCCESS in 53m 05s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/6/check-ibm-tempest-dsvm-postgres-full/436f255 : SUCCESS in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0554d239f28657dd23883975ad4ca2b83955f5a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-04-21 18:54:15.000000000","message":"Patch Set 6:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/70/162470/6/15436/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"c67a97e9f4aea683740ffc4c046e6345b3327ad6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-04-21 18:55:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/70/162470/6/passonly/dsvm-tempest-xen/34dd263 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":6},{"id":"7daec444e2844b7b84ba8661b4cc23b1200fa018","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-21 19:12:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/70/162470/6/check/gate-nova-pep8/44970f5/ : SUCCESS in 10m 44s\n- gate-nova-docs http://docs-draft.openstack.org/70/162470/6/check/gate-nova-docs/febab9a//doc/build/html/ : SUCCESS in 11m 10s\n- gate-nova-python27 http://logs.openstack.org/70/162470/6/check/gate-nova-python27/2cacbf3/ : SUCCESS in 16m 09s\n- check-tempest-dsvm-full http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-full/0f0f7cf/ : SUCCESS in 40m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-postgres-full/2d1af43/ : SUCCESS in 40m 45s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-neutron-full/dc5c82b/ : SUCCESS in 57m 38s\n- check-grenade-dsvm http://logs.openstack.org/70/162470/6/check/check-grenade-dsvm/d9bcd63/ : FAILURE in 14m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/70/162470/6/check/gate-tempest-dsvm-large-ops/458fe2b/ : SUCCESS in 21m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/70/162470/6/check/gate-tempest-dsvm-neutron-large-ops/a8a7eb9/ : SUCCESS in 19m 58s\n- check-devstack-dsvm-cells http://logs.openstack.org/70/162470/6/check/check-devstack-dsvm-cells/b2ef1ad/ : SUCCESS in 22m 26s\n- gate-nova-tox-functional http://logs.openstack.org/70/162470/6/check/gate-nova-tox-functional/a9b0f00/ : SUCCESS in 14m 47s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/70/162470/6/check/check-grenade-dsvm-partial-ncpu/d75c2d2/ : FAILURE in 14m 49s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-ironic-pxe_ssh/682bd66/ : SUCCESS in 31m 47s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-nova-v21-full/11230cb/ : SUCCESS in 47m 13s\n- check-tempest-dsvm-cells http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-cells/c39c002/ : FAILURE in 37m 16s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-full-ceph/1c2b6c8/ : SUCCESS in 40m 29s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/70/162470/6/check/check-tempest-dsvm-multinode-full/0ce82e1/ : SUCCESS in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c4c3047bd338c7f0a3c56ee193bcd3555cc25829","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-04-21 20:37:40.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/162470/6/ : FAILURE in 1h 44m 45s","accounts_in_message":[],"_revision_number":6},{"id":"fc20948ac1b3e19b985d7777d230a4eab342d99a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-04-25 09:32:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/162470/5 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/162470/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"1f95260d04ee9dac4369aec8b9012f894d757437","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-04-25 10:08:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/162470/6 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/162470/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"36355067c9dfb52466f60e972ea2a7988a0e1246","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2015-04-29 07:54:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d571bfd826c6087e1cfcad892e17d3ce1f93e30f","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-05-07 20:23:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"862fc751c3681e4d46b0b0c9d872b9f146e95bbb","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2015-05-09 12:21:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"22b38179eece1ef58f68d51d03a75ef523a87410","author":{"_account_id":13734,"name":"Nell Jerram","email":"nell@tigera.io","username":"neiljerram"},"date":"2015-05-12 17:22:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1ead60d9e35f5126f49066a9af252069e96adb86","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2015-05-22 08:26:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"00d8d6655276b83ce2083b0c9a8ed748cadddc29","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-05-22 16:53:48.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"116dc51252abdc4864f1866870de683a92dbc620","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-05-26 19:51:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5803633a83babd83e2d635de9afe636d8c2a49f3","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-05-26 19:52:27.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"476efd8277990b68afd38f1c728a77b8defe9def","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-05-26 20:03:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Check-MLNX-Nova-ML2-Sriov-driver-port http://144.76.193.39/ci-artifacts/162470/7/Check-MLNX-Nova-ML2-Sriov-driver-port : FAILURE in 9m 16s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"3f7fa09ca80787a7f962cb476c7f4f7962b98f90","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-26 20:03:52.000000000","message":"Patch Set 7:\n\n* pci-test http://198.175.100.33/162470/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"7b093feef3eef6ab295856cf13bf48f5cf237ceb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 20:45:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/70/162470/7/check/gate-nova-pep8/be94c1e/ : SUCCESS in 8m 28s\n- gate-nova-docs http://docs-draft.openstack.org/70/162470/7/check/gate-nova-docs/4960347//doc/build/html/ : SUCCESS in 7m 51s\n- gate-nova-python27 http://logs.openstack.org/70/162470/7/check/gate-nova-python27/3ea1903/ : FAILURE in 12m 41s\n- check-tempest-dsvm-full http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-full/7fb3158/ : SUCCESS in 49m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-postgres-full/62278e7/ : SUCCESS in 35m 11s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-neutron-full/097ee54/ : SUCCESS in 43m 03s\n- check-grenade-dsvm http://logs.openstack.org/70/162470/7/check/check-grenade-dsvm/56730e5/ : SUCCESS in 36m 54s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/70/162470/7/check/gate-tempest-dsvm-large-ops/13d5861/ : SUCCESS in 31m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/70/162470/7/check/gate-tempest-dsvm-neutron-large-ops/aaff73d/ : SUCCESS in 25m 21s\n- check-devstack-dsvm-cells http://logs.openstack.org/70/162470/7/check/check-devstack-dsvm-cells/f78e6df/ : FAILURE in 20m 03s\n- gate-nova-tox-functional http://logs.openstack.org/70/162470/7/check/gate-nova-tox-functional/fa70437/ : SUCCESS in 11m 14s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/70/162470/7/check/check-grenade-dsvm-partial-ncpu/7c22a62/ : FAILURE in 30m 56s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-ironic-pxe_ssh/acc591d/ : SUCCESS in 28m 27s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-nova-v21-full/821370b/ : SUCCESS in 36m 32s\n- check-tempest-dsvm-cells http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-cells/ccefab1/ : FAILURE in 33m 54s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-full-ceph/8575240/ : SUCCESS in 41m 26s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/70/162470/7/check/check-tempest-dsvm-multinode-full/b96adc4/ : FAILURE in 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"59ddce8a1453dd589e4186f233faf8582d0de4d5","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-05-26 20:45:51.000000000","message":"Patch Set 7:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- check-grenade-dsvm-partial-ncpu: unrecognized error\n- gate-nova-python27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":7},{"id":"931d4dbe83584f593d4db6e940e1a03e77cad4fa","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-05-26 20:51:17.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/59a322b7e97747238d4ddec8d836c881/index.html : SUCCESS in 17m 42s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7fdeca943f59425296d5c676eb30f475/index.html : SUCCESS in 27m 02s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dc592ea7acef434391159625971816ff/index.html : SUCCESS in 18m 15s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/7/check/real-db-upgrade_nova_percona_user_001:th-percona/9d49fc7faad644478027f66aa0abff46/index.html : SUCCESS in 22m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3eacb92ceab04f7b81b10d2cb518d7ef/index.html : SUCCESS in 21m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/7/check/real-db-upgrade_nova_percona_user_002:th-percona/e580e8905d8c45b293c793a094b25de0/index.html : SUCCESS in 20m 55s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"4bdd6f9599ba23217383c161c58c3c8d20fce9a1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-05-26 21:01:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/70/162470/7/check/dsvm-tempest-xen/1204c28 : SUCCESS in 1h 09m 03s","accounts_in_message":[],"_revision_number":7},{"id":"6d61f1df00a8bca4d45223d7eb3bc17531a4eb5f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-05-26 21:04:57.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/7/check-ibm-tempest-dsvm-full/5061db8 : SUCCESS in 1h 12m 39s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/7/check-ibm-tempest-dsvm-postgres-full/e595244 : SUCCESS in 1h 04m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2f89a74bc42c4e3eadd22b0b30c33511a43b05a6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-05-26 21:08:57.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/162470/7/ : FAILURE in 1h 16m 28s","accounts_in_message":[],"_revision_number":7},{"id":"224c35e1edfc09866d082b26d9c9c8af9bb5799d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-27 04:13:14.000000000","message":"Patch Set 7:\n\n* pci-test http://198.175.100.33/162470/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"9a26325595b0edcde799f610aff744ec81639582","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-05-27 16:18:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"20f21952ed69e7254c1bcf68c5310fe3474e4c1a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-27 16:33:03.000000000","message":"Patch Set 8:\n\n* pci-test http://198.175.100.33/162470/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"e119cbbb2ac9b990c27ae7149b2846e2c4f0375d","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-05-27 16:35:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"17e2297b6807b95ce1892a10b8ee269de3faaf28","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-27 16:48:01.000000000","message":"Patch Set 9:\n\n* pci-test http://198.175.100.33/162470/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"dc34234890d6670ddda0a82ff1f9f2c52c231043","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-27 17:34:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/70/162470/9/check/gate-nova-pep8/72a40a4/ : SUCCESS in 10m 51s\n- gate-nova-docs http://docs-draft.openstack.org/70/162470/9/check/gate-nova-docs/d6e4778//doc/build/html/ : SUCCESS in 7m 58s\n- gate-nova-python27 http://logs.openstack.org/70/162470/9/check/gate-nova-python27/6c57742/ : FAILURE in 11m 57s\n- check-tempest-dsvm-full http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-full/3d8efd8/ : SUCCESS in 46m 18s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-postgres-full/ecfe78e/ : SUCCESS in 49m 01s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-neutron-full/260ecee/ : SUCCESS in 56m 13s\n- check-grenade-dsvm http://logs.openstack.org/70/162470/9/check/check-grenade-dsvm/adccd42/ : SUCCESS in 40m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/70/162470/9/check/gate-tempest-dsvm-large-ops/46755eb/ : SUCCESS in 28m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/70/162470/9/check/gate-tempest-dsvm-neutron-large-ops/fed0774/ : SUCCESS in 19m 09s\n- check-devstack-dsvm-cells http://logs.openstack.org/70/162470/9/check/check-devstack-dsvm-cells/e6b0b6d/ : FAILURE in 21m 40s\n- gate-nova-tox-functional http://logs.openstack.org/70/162470/9/check/gate-nova-tox-functional/da4cabe/ : SUCCESS in 14m 46s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/70/162470/9/check/check-grenade-dsvm-partial-ncpu/db0a7bc/ : FAILURE in 36m 31s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-ironic-pxe_ssh/b035139/ : SUCCESS in 30m 24s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-nova-v21-full/104c20a/ : FAILURE in 37m 10s\n- check-tempest-dsvm-cells http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-cells/6415615/ : FAILURE in 28m 12s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-full-ceph/03f24e3/ : SUCCESS in 46m 54s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/70/162470/9/check/check-tempest-dsvm-multinode-full/2858881/ : FAILURE in 48m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4e63845b4eb7df654722b8ed19056529eb21aa87","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-05-27 17:35:09.000000000","message":"Patch Set 9:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- check-grenade-dsvm-partial-ncpu: unrecognized error\n- check-tempest-dsvm-nova-v21-full: unrecognized error\n- gate-nova-python27: https://bugs.launchpad.net/bugs/1459021\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"b7f70b789d4a8c66cfc7b2af11273c81e15886bf","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-05-27 17:35:13.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"19ca3749e9d655ccbb32911076922549578e8abf","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-05-27 17:38:56.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/162470/9/ : FAILURE in 1h 03m 07s","accounts_in_message":[],"_revision_number":9},{"id":"05afd95aa7fad8304cd3e8767a7b3a45d2e31ac4","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-05-27 17:44:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/70/162470/9/check/dsvm-tempest-xen/dbeec1f : FAILURE in 1h 08m 35s","accounts_in_message":[],"_revision_number":9},{"id":"ca19d1592a58a529b0cb0355acd55e74f9adc541","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-05-27 18:04:47.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/9/check-ibm-tempest-dsvm-full/0f43ac0 : SUCCESS in 1h 10m 02s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/162470/9/check-ibm-tempest-dsvm-postgres-full/3372e61 : SUCCESS in 1h 04m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d8733ac055cf6d1287790c7ec345b72325b3e0fa","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-05-28 01:26:01.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/acef5a70f4ca4413b6e984ba08e20dce/index.html : SUCCESS in 20m 39s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/279c8aa3d0004b9e83978ff723a5d4a6/index.html : SUCCESS in 25m 15s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fed7c08a19b143a6bc257ec83847c3bd/index.html : SUCCESS in 19m 36s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/9/check/real-db-upgrade_nova_percona_user_001:th-percona/63fa857d81d140de9a9bd4e729f7c3bf/index.html : SUCCESS in 26m 48s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/70/162470/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5471df3ea8094a80a0a733e9751fe5a8/index.html : SUCCESS in 20m 45s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/70/162470/9/check/real-db-upgrade_nova_percona_user_002:th-percona/39a7130dcb9d4dfebc429a3baa7ff50d/index.html : SUCCESS in 19m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"c55d7d9f1b6b68e2c3762274c37e0b1b7f2cd993","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-05-28 03:29:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/70/162470/8/check/dsvm-tempest-xen/a564eae : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":8},{"id":"a586db29f45e5a7af3477a8fa4fc66993edf7c1c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-05-28 03:29:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/70/162470/9/check/dsvm-tempest-xen/c504c23 : SUCCESS in 1h 07m 06s","accounts_in_message":[],"_revision_number":9},{"id":"eec83d35e32458721c484687deac8ecf51c4182b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-05-28 15:36:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/162470/9 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/162470/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"91243ad6b3cc415f53feca8ee97b1136bf0e50de","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2015-05-31 08:06:44.000000000","message":"Patch Set 9:\n\n(2 comments)\n\nfew comments inline","accounts_in_message":[],"_revision_number":9},{"id":"7d39ed460537970cde198068bab71243958b3853","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2015-06-02 16:19:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"48cb88ed8d5692869a3eb1468f0905e4cb3776c8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2015-06-04 17:45:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a15199548b70993d83b8e056c6fbf9f99553ff8e","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2015-06-25 14:49:46.000000000","message":"Abandoned\n\nMoving towards a different approach","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"487acfb343cda855169d37be126598459b36ec84","revisions":{"493e2b42c52ce4e253f9669885f2591c743a4fae":{"kind":"REWORK","_number":1,"created":"2015-03-08 13:21:45.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/1"}}},"commit":{"parents":[{"commit":"4103a86f30bf7da3e4616d9d769213b128f0946d","subject":"Merge \"Replace RPC topic-based service queries in nova/api with binary-based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4103a86f30bf7da3e4616d9d769213b128f0946d"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:21:42.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/493e2b42c52ce4e253f9669885f2591c743a4fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/493e2b42c52ce4e253f9669885f2591c743a4fae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4103a86f30bf7da3e4616d9d769213b128f0946d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9f6315c7ba5104693663c122b63f88d959a795d6":{"kind":"REWORK","_number":2,"created":"2015-03-08 21:48:43.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/2"}}},"commit":{"parents":[{"commit":"4103a86f30bf7da3e4616d9d769213b128f0946d","subject":"Merge \"Replace RPC topic-based service queries in nova/api with binary-based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4103a86f30bf7da3e4616d9d769213b128f0946d"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 21:48:29.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f6315c7ba5104693663c122b63f88d959a795d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f6315c7ba5104693663c122b63f88d959a795d6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4103a86f30bf7da3e4616d9d769213b128f0946d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"00c042b2a03a9b80f8de7deacbd1528da6fe1f3b":{"kind":"REWORK","_number":3,"created":"2015-03-25 19:49:42.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/3"}}},"commit":{"parents":[{"commit":"cdad99ea4a14c9d34efed917628fd4e68634c8b9","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdad99ea4a14c9d34efed917628fd4e68634c8b9"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-25 19:48:53.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00c042b2a03a9b80f8de7deacbd1528da6fe1f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00c042b2a03a9b80f8de7deacbd1528da6fe1f3b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cdad99ea4a14c9d34efed917628fd4e68634c8b9","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"42c359088f6c97972dd5ee8d897ac9dd00578246":{"kind":"REWORK","_number":4,"created":"2015-03-25 19:55:34.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/4"}}},"commit":{"parents":[{"commit":"cdad99ea4a14c9d34efed917628fd4e68634c8b9","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdad99ea4a14c9d34efed917628fd4e68634c8b9"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-25 19:54:45.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/42c359088f6c97972dd5ee8d897ac9dd00578246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/42c359088f6c97972dd5ee8d897ac9dd00578246"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cdad99ea4a14c9d34efed917628fd4e68634c8b9","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"28dd7051f70e4c6cd97357272d7f360f42a84662":{"kind":"REWORK","_number":5,"created":"2015-04-21 17:33:41.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/5"}}},"commit":{"parents":[{"commit":"4f7d78da637cce8326c7f64522469ef95bee5695","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f7d78da637cce8326c7f64522469ef95bee5695"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-04-21 17:32:30.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28dd7051f70e4c6cd97357272d7f360f42a84662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28dd7051f70e4c6cd97357272d7f360f42a84662"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4f7d78da637cce8326c7f64522469ef95bee5695","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":3,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"25c9ac0f7e11bf34550cf5f0c71895a7d0463f95":{"kind":"REWORK","_number":6,"created":"2015-04-21 17:49:57.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/6"}}},"commit":{"parents":[{"commit":"4f7d78da637cce8326c7f64522469ef95bee5695","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f7d78da637cce8326c7f64522469ef95bee5695"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-04-21 17:49:44.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25c9ac0f7e11bf34550cf5f0c71895a7d0463f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25c9ac0f7e11bf34550cf5f0c71895a7d0463f95"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4f7d78da637cce8326c7f64522469ef95bee5695","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":3,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"1d2de69cd0e069ab386569bafd94b93b67103d31":{"kind":"REWORK","_number":7,"created":"2015-05-26 19:51:51.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/7"}}},"commit":{"parents":[{"commit":"0dabbccb46b090cc50b8d15d3da3753a0b216280","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dabbccb46b090cc50b8d15d3da3753a0b216280"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-05-26 19:50:14.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nWIP TODO: Add configuration variable for script locations, validation\nfor path info, stderr logging and error code handling.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2de69cd0e069ab386569bafd94b93b67103d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2de69cd0e069ab386569bafd94b93b67103d31"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0dabbccb46b090cc50b8d15d3da3753a0b216280","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"7d5877c013ec1b1970975814450c279d8691c26e":{"kind":"REWORK","_number":8,"created":"2015-05-27 16:18:27.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/8"}}},"commit":{"parents":[{"commit":"0dabbccb46b090cc50b8d15d3da3753a0b216280","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dabbccb46b090cc50b8d15d3da3753a0b216280"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-05-27 16:17:36.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nWIP note: this is a proof-of-concept that is tracking a spec proposal\nthat is not yet approved.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d5877c013ec1b1970975814450c279d8691c26e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d5877c013ec1b1970975814450c279d8691c26e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0dabbccb46b090cc50b8d15d3da3753a0b216280","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"487acfb343cda855169d37be126598459b36ec84":{"kind":"REWORK","_number":9,"created":"2015-05-27 16:35:15.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/70/162470/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/70/162470/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/70/162470/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/70/162470/9"}}},"commit":{"parents":[{"commit":"0dabbccb46b090cc50b8d15d3da3753a0b216280","subject":"WIP: Refactor Nova VIF model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dabbccb46b090cc50b8d15d3da3753a0b216280"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-03-08 13:17:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2015-05-27 16:34:58.000000000","tz":-150},"subject":"WIP: Add support for VIF plugin script to Nova","message":"WIP: Add support for VIF plugin script to Nova\n\nThis change introduces support for handling new VIF detail field,\nVIF_PLUGIN_SCRIPT. This field will be populated by Neutron and refer to\na back-end specific script that will be called from the virtualization\ndriver when the system is ready to connect the guest to a L2 network.\n\nWIP note: this is a proof-of-concept that is tracking a spec proposal\nthat is not yet approved.\n\nChange-Id: I3b27ba441c9c3e84c8f31b03ab630d6a1c49abb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/487acfb343cda855169d37be126598459b36ec84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/487acfb343cda855169d37be126598459b36ec84"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0dabbccb46b090cc50b8d15d3da3753a0b216280","is_merged_in_target_branch":false,"change_id":"Ia0f463e08dc03afef11332aa6d1c643415e49d93","change_number":167287,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
