)]}'
{"id":"openstack%2Fnova~165425","triplet_id":"openstack%2Fnova~master~I1993cca4cf5cf6eec6370597b37d4f53127d67a0","project":"openstack/nova","branch":"master","topic":"bug/get_version","hashtags":[],"change_id":"I1993cca4cf5cf6eec6370597b37d4f53127d67a0","subject":"remove duplicate calls to cfg.get()","status":"MERGED","created":"2015-03-18 14:04:48.000000000","updated":"2015-03-18 19:14:00.000000000","submitted":"2015-03-18 19:13:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"89d14ce5aabfb5a2bb717500c52ebe50b68a90cf","_number":165425,"virtual_id_number":165425,"owner":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2015-03-18 15:16:10.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-03-18 15:06:12.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":2,"date":"2015-03-18 19:13:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-03-18 15:13:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":1,"date":"2015-03-18 14:30:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-03-18 15:34:54.000000000","_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":2,"date":"2015-03-18 15:17:19.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2015-03-18 15:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-03-18 15:34:54.000000000","_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-03-18 14:30:08.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-03-18 15:06:12.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-03-18 15:13:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-03-18 15:16:10.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-03-18 15:17:19.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2015-03-18 15:25:02.000000000","updated_by":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"reviewer":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"state":"REVIEWER"},{"updated":"2015-03-18 15:34:54.000000000","updated_by":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"reviewer":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"state":"REVIEWER"},{"updated":"2015-03-18 19:13:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8427bdcf393636a6faed7f2d8213cd028f5fdaf0","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2015-03-18 14:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20677a6264d1b13907ebb7f8d8dd076eb35f698b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-03-18 14:30:08.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/25/165425/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2e081e96fb174fb59d22a13b155e0b2d/index.html : SUCCESS in 17m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/25/165425/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2a21e5370428414cb3566853f48fc179/index.html : SUCCESS in 24m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/25/165425/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1485ef088d894d35971487a86d475727/index.html : SUCCESS in 20m 30s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/25/165425/1/check/real-db-upgrade_nova_percona_user_001:th-percona/f004a01f578a4cb2b446e3deccdeb9da/index.html : SUCCESS in 21m 26s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/25/165425/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/581af7a74edd4dd0ad6692f852c5e09b/index.html : SUCCESS in 20m 54s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/25/165425/1/check/real-db-upgrade_nova_percona_user_002:th-percona/7033ce769c614bcfb7d9dd1c7b454ae5/index.html : SUCCESS in 20m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2a9f75eb984f280eb7bcc061758faeee7cff1dc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-03-18 15:06:12.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/165425/1/ : Test run failed. in 3s","accounts_in_message":[],"_revision_number":1},{"id":"e45a4aa9eb45ac480d6cad46513d727958675b8f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-03-18 15:13:55.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/165425/1/check-ibm-tempest-dsvm-full/35de10f : SUCCESS in 1h 08m 29s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/165425/1/check-ibm-tempest-dsvm-postgres-full/e2730a2 : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c79cbe3ac14d829bb2e9a20ac0c29050b6932c4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-03-18 15:16:10.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/25/165425/1/13823/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"a10cf6cdf818c107949610a085975c39759e0b3d","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-03-18 15:17:19.000000000","message":"Patch Set 1: Code-Review+2\n\nNo idea what I was smoking when I did that. The first 3 versions of the original review were sane, but the last one wasn\u0027t. No idea how that got approved either :-)","accounts_in_message":[],"_revision_number":1},{"id":"7b4a7c904c869af389905821a9e43bd1ccfabe2b","author":{"_account_id":8247,"name":"Thang Pham","email":"thang.g.pham@gmail.com","username":"thang.pham"},"date":"2015-03-18 15:25:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d838428f4d5be4e7d7d7224e1cbd3db5c481e035","author":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"date":"2015-03-18 15:34:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83000968042c63a5f5c40f368cec82a2a0ba4634","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-18 15:38:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/25/165425/1/check/gate-nova-pep8/5c55258/ : SUCCESS in 11m 25s\n- gate-nova-docs http://docs-draft.openstack.org/25/165425/1/check/gate-nova-docs/b283b62//doc/build/html/ : SUCCESS in 9m 36s\n- gate-nova-python27 http://logs.openstack.org/25/165425/1/check/gate-nova-python27/d40a024/ : SUCCESS in 14m 34s\n- check-tempest-dsvm-full http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-full/15e42c5/ : SUCCESS in 43m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-postgres-full/0e1885b/ : SUCCESS in 43m 41s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-neutron-full/413f6d5/ : SUCCESS in 55m 02s\n- check-grenade-dsvm http://logs.openstack.org/25/165425/1/check/check-grenade-dsvm/b547879/ : SUCCESS in 40m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/25/165425/1/check/gate-tempest-dsvm-large-ops/cb7332a/ : SUCCESS in 26m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/25/165425/1/check/gate-tempest-dsvm-neutron-large-ops/3992a8b/ : SUCCESS in 25m 25s\n- check-devstack-dsvm-cells http://logs.openstack.org/25/165425/1/check/check-devstack-dsvm-cells/9dc11ee/ : SUCCESS in 24m 25s\n- gate-nova-tox-functional http://logs.openstack.org/25/165425/1/check/gate-nova-tox-functional/054177b/ : SUCCESS in 11m 57s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/25/165425/1/check/check-grenade-dsvm-partial-ncpu/ab5c310/ : SUCCESS in 48m 14s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-ironic-pxe_ssh/171fd08/ : SUCCESS in 36m 47s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-nova-v21-full/a7ef0a6/ : SUCCESS in 45m 24s\n- check-tempest-dsvm-cells http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-cells/3072978/ : FAILURE in 34m 24s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/25/165425/1/check/check-tempest-dsvm-full-ceph/aa2976e/ : SUCCESS in 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"496905716cc85072167a6a86d3f3c67b1b475029","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-18 15:38:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"78ce2cba1a3a9862245a4f6116732ae927d9b133","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-18 19:13:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/25/165425/1/gate/gate-nova-docs/78c0484//doc/build/html/ : SUCCESS in 11m 51s\n- gate-nova-pep8 http://logs.openstack.org/25/165425/1/gate/gate-nova-pep8/666fef1/ : SUCCESS in 8m 10s\n- gate-nova-python27 http://logs.openstack.org/25/165425/1/gate/gate-nova-python27/b153420/ : SUCCESS in 13m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/25/165425/1/gate/gate-tempest-dsvm-full/fcd28be/ : SUCCESS in 40m 57s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/25/165425/1/gate/gate-tempest-dsvm-postgres-full/c2dba74/ : SUCCESS in 46m 51s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/25/165425/1/gate/gate-tempest-dsvm-neutron-full/29a51a1/ : SUCCESS in 1h 07m 13s\n- gate-grenade-dsvm http://logs.openstack.org/25/165425/1/gate/gate-grenade-dsvm/2244e14/ : SUCCESS in 43m 54s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/25/165425/1/gate/gate-tempest-dsvm-large-ops/5ea16b1/ : SUCCESS in 28m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/25/165425/1/gate/gate-tempest-dsvm-neutron-large-ops/8487943/ : SUCCESS in 28m 44s\n- gate-nova-tox-functional http://logs.openstack.org/25/165425/1/gate/gate-nova-tox-functional/bb0cc47/ : SUCCESS in 11m 25s\n- gate-devstack-dsvm-cells http://logs.openstack.org/25/165425/1/gate/gate-devstack-dsvm-cells/e01cb26/ : SUCCESS in 22m 03s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/25/165425/1/gate/gate-grenade-dsvm-partial-ncpu/285b823/ : SUCCESS in 40m 18s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/25/165425/1/gate/gate-tempest-dsvm-nova-v21-full/63fac82/ : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":1},{"id":"41998d9f03c526097448603575dc80fbc38ec5f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-18 19:14:00.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce68b7f025d1ee25a58a093adf462b4b77fb0ad4","revisions":{"ce68b7f025d1ee25a58a093adf462b4b77fb0ad4":{"kind":"REWORK","_number":1,"created":"2015-03-18 14:04:48.000000000","uploader":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"ref":"refs/changes/25/165425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/25/165425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/25/165425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/25/165425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/25/165425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/25/165425/1"}}},"commit":{"parents":[{"commit":"3abdda5adbce3fb9f237de6f2471d7d90856a9e0","subject":"Merge \"Removed host_id check in ServersController.update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3abdda5adbce3fb9f237de6f2471d7d90856a9e0"}]}],"author":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2015-03-18 14:03:26.000000000","tz":-240},"committer":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2015-03-18 14:03:26.000000000","tz":-240},"subject":"remove duplicate calls to cfg.get()","message":"remove duplicate calls to cfg.get()\n\nIn nova/version.py, we were making duplicate calls to cfg.get(), like\nthis:\n\n  NOVA_VENDOR \u003d cfg.get(\"Nova\", \"vendor\")\n  if cfg.has_option(\"Nova\", \"vendor\"):\n    NOVA_VENDOR \u003d cfg.get(\"Nova\", \"vendor\")\n\nThis removes the \"unprotected\" calls.\n\nChange-Id: I1993cca4cf5cf6eec6370597b37d4f53127d67a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce68b7f025d1ee25a58a093adf462b4b77fb0ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce68b7f025d1ee25a58a093adf462b4b77fb0ad4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3abdda5adbce3fb9f237de6f2471d7d90856a9e0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
