)]}'
{"id":"openstack%2Fnova~171390","triplet_id":"openstack%2Fnova~master~I326ffaaab941e9fb897a03da3512e4cb481e4a3e","project":"openstack/nova","branch":"master","topic":"bug/1249065","hashtags":[],"change_id":"I326ffaaab941e9fb897a03da3512e4cb481e4a3e","subject":"Allow refreshing network info_cache in get_nw_info_for_instance","status":"ABANDONED","created":"2015-04-07 21:29:45.000000000","updated":"2015-06-16 16:31:00.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48f7b4090b47f64e3de47edc14f6c582fcd70f33","_number":171390,"virtual_id_number":171390,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2015-04-25 11:27:32.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-04-27 13:31:12.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2015-04-26 06:45:20.000000000","_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"date":"2015-04-21 21:56:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"date":"2015-04-21 22:35:22.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"date":"2015-04-21 21:56:31.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"date":"2015-04-22 13:33:50.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2015-04-27 03:36:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"date":"2015-04-26 06:10:49.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2015-04-22 13:34:07.000000000","_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-04-28 17:52:58.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-04-07 22:39:46.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 21:56:31.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 21:56:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 22:35:22.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-04-22 13:34:07.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2015-04-25 11:27:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-26 06:10:49.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2015-04-26 06:45:20.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2015-04-27 03:36:20.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-27 13:31:12.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"467626a8586c24fb884ba3efdc09f91118d24cab","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-07 21:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11b8829734a7a7280006fbeec6f555ae19d6c342","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-07 21:30:30.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds tests.","accounts_in_message":[],"_revision_number":1},{"id":"f679d4872308ab3653893c2d92866473faea5fe4","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-07 21:54:44.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/90/171390/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9176fe4de3aa42968984d5c6510a1dd1/index.html : SUCCESS in 21m 40s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/90/171390/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/aed0825d14e54254ac2b9a8552a8afa7/index.html : SUCCESS in 23m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/90/171390/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d3818c97d5e843cbba12ed1590831b6f/index.html : SUCCESS in 21m 35s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/90/171390/1/check/real-db-upgrade_nova_percona_user_001:th-percona/94d245347a3c49e28d93bba3443fb405/index.html : SUCCESS in 24m 15s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/90/171390/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3f25b2bd5aab409daea2545dd7a1d86e/index.html : SUCCESS in 18m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/90/171390/1/check/real-db-upgrade_nova_percona_user_002:th-percona/4e9ef72db0fc46039bdf6c7fb6109d0b/index.html : SUCCESS in 18m 25s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"1a016efffbafab6e1e3654b696880456dafe1c96","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-07 22:12:49.000000000","message":"Patch Set 1:\n\nBuild Successful. Logs at http://198.175.100.33/171390/1","accounts_in_message":[],"_revision_number":1},{"id":"cc9e40558ab3584ba8014faca6f5563f5e39ca11","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-04-07 22:39:46.000000000","message":"Patch Set 1: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/90/171390/1/14714/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"6fab826224b81d252ccad621b5641ab7a7a09173","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-07 23:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/171390/1/check/gate-nova-pep8/53f153e/ : SUCCESS in 8m 08s\n- gate-nova-docs http://docs-draft.openstack.org/90/171390/1/check/gate-nova-docs/3470e69//doc/build/html/ : SUCCESS in 8m 24s\n- gate-nova-python27 http://logs.openstack.org/90/171390/1/check/gate-nova-python27/550625d/ : SUCCESS in 16m 21s\n- check-tempest-dsvm-full http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-full/19b1ab4/ : SUCCESS in 41m 13s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-postgres-full/eaccdb8/ : SUCCESS in 43m 53s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-neutron-full/9c8eec4/ : SUCCESS in 1h 08m 36s\n- check-grenade-dsvm http://logs.openstack.org/90/171390/1/check/check-grenade-dsvm/c558a6e/ : SUCCESS in 38m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/90/171390/1/check/gate-tempest-dsvm-large-ops/cb84b02/ : SUCCESS in 21m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/90/171390/1/check/gate-tempest-dsvm-neutron-large-ops/c0430c4/ : SUCCESS in 21m 39s\n- check-devstack-dsvm-cells http://logs.openstack.org/90/171390/1/check/check-devstack-dsvm-cells/46f4f19/ : SUCCESS in 18m 56s\n- gate-nova-tox-functional http://logs.openstack.org/90/171390/1/check/gate-nova-tox-functional/7d7f9f0/ : SUCCESS in 11m 11s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/90/171390/1/check/check-grenade-dsvm-partial-ncpu/d902a25/ : SUCCESS in 40m 09s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-ironic-pxe_ssh/e582812/ : SUCCESS in 30m 14s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-nova-v21-full/8975ed7/ : SUCCESS in 42m 03s\n- check-tempest-dsvm-cells http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-cells/bc773ff/ : SUCCESS in 22m 39s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-full-ceph/3889f5d/ : FAILURE in 38m 01s (non-voting)\n- check-tempest-dsvm-aiopcpu-full http://logs.openstack.org/90/171390/1/check/check-tempest-dsvm-aiopcpu-full/a36ad80/ : FAILURE in 57m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ea14e7f6f320a96c6df78f78d0613af140fa501","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-07 23:42:23.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/171390/1/check-ibm-tempest-dsvm-full/80f82f0 : SUCCESS in 51m 49s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/171390/1/check-ibm-tempest-dsvm-postgres-full/28cc0cf : ABORTED in 2h 08m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"423cb9e7b7477058812d3a327ab9d8d38803d777","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-21 21:43:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df33b7b9507391c8bda371acdb322ac1bfe3f0aa","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-21 21:56:31.000000000","message":"Patch Set 2:\n\n* pci-test http://198.175.100.33/171390/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e8bbc3af1dbb59ba11478148335ac3ebf0bd9897","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-21 21:56:43.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/171390/2/check-ibm-tempest-dsvm-full/3a0fc8c : FAILURE in 10m 34s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/171390/2/check-ibm-tempest-dsvm-postgres-full/c73cee6 : FAILURE in 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdc18565942184d3db42870af0ac92705da60f7a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-21 22:32:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/90/171390/2/check/gate-nova-pep8/4f937da/ : FAILURE in 2m 54s\n- gate-nova-docs http://logs.openstack.org/90/171390/2/check/gate-nova-docs/00b3213/ : FAILURE in 3m 05s\n- gate-nova-python27 http://logs.openstack.org/90/171390/2/check/gate-nova-python27/0deba84/ : FAILURE in 2m 57s\n- check-tempest-dsvm-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-full/746c69b/ : FAILURE in 6m 48s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-postgres-full/e0238e2/ : FAILURE in 7m 31s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-neutron-full/9d16da4/ : FAILURE in 8m 30s\n- check-grenade-dsvm http://logs.openstack.org/90/171390/2/check/check-grenade-dsvm/734ef0f/ : FAILURE in 16m 27s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/90/171390/2/check/gate-tempest-dsvm-large-ops/92bdca6/ : FAILURE in 7m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/90/171390/2/check/gate-tempest-dsvm-neutron-large-ops/2a43a09/ : FAILURE in 7m 40s\n- check-devstack-dsvm-cells http://logs.openstack.org/90/171390/2/check/check-devstack-dsvm-cells/31d0f2a/ : FAILURE in 7m 19s\n- gate-nova-tox-functional http://logs.openstack.org/90/171390/2/check/gate-nova-tox-functional/22e6932/ : FAILURE in 3m 55s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/90/171390/2/check/check-grenade-dsvm-partial-ncpu/5053dc5/ : FAILURE in 13m 31s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-ironic-pxe_ssh/3af909b/ : FAILURE in 6m 50s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-nova-v21-full/230afd7/ : FAILURE in 7m 36s\n- check-tempest-dsvm-cells http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-cells/5a40150/ : FAILURE in 7m 28s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-full-ceph/87fcabc/ : FAILURE in 6m 21s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-multinode-full/c60d85b/ : FAILURE in 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d479784c1432cc3a6b1f8120d817fa636790125","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-04-21 22:35:22.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: unrecognized error\n- check-grenade-dsvm-partial-ncpu: https://bugs.launchpad.net/bugs/1446847\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1446847\n- check-tempest-dsvm-full-ceph: unrecognized error\n- check-tempest-dsvm-full: unrecognized error\n- check-tempest-dsvm-ironic-pxe_ssh: unrecognized error\n- check-tempest-dsvm-neutron-full: unrecognized error\n- check-tempest-dsvm-nova-v21-full: unrecognized error\n- check-tempest-dsvm-postgres-full: unrecognized error\n- gate-nova-docs: unrecognized error\n- gate-nova-pep8: unrecognized error\n- gate-nova-python27: unrecognized error\n- gate-nova-tox-functional: unrecognized error\n- gate-tempest-dsvm-large-ops: unrecognized error\n- gate-tempest-dsvm-neutron-large-ops: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"5ffaad53c1fcdd52d06e95c285092803716fae0f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-22 13:33:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"28617decffc9190a4a83a772b0faadc9ff7b63a5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 15:14:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/171390/2/check/gate-nova-pep8/bd80d37/ : SUCCESS in 11m 05s\n- gate-nova-docs http://docs-draft.openstack.org/90/171390/2/check/gate-nova-docs/1c8acf1//doc/build/html/ : SUCCESS in 8m 57s\n- gate-nova-python27 http://logs.openstack.org/90/171390/2/check/gate-nova-python27/1cba966/ : SUCCESS in 12m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-full/6f35acf/ : SUCCESS in 40m 52s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-postgres-full/b4bbcdf/ : SUCCESS in 42m 03s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-neutron-full/bd257ba/ : SUCCESS in 49m 59s\n- check-grenade-dsvm http://logs.openstack.org/90/171390/2/check/check-grenade-dsvm/31862f6/ : SUCCESS in 40m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/90/171390/2/check/gate-tempest-dsvm-large-ops/fcc076c/ : SUCCESS in 20m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/90/171390/2/check/gate-tempest-dsvm-neutron-large-ops/51e0530/ : SUCCESS in 20m 09s\n- check-devstack-dsvm-cells http://logs.openstack.org/90/171390/2/check/check-devstack-dsvm-cells/f26e72f/ : SUCCESS in 18m 51s\n- gate-nova-tox-functional http://logs.openstack.org/90/171390/2/check/gate-nova-tox-functional/5e4498b/ : SUCCESS in 11m 16s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/90/171390/2/check/check-grenade-dsvm-partial-ncpu/1c49f76/ : SUCCESS in 38m 40s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-ironic-pxe_ssh/4fe6ceb/ : SUCCESS in 30m 40s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-nova-v21-full/8490b7b/ : SUCCESS in 38m 40s\n- check-tempest-dsvm-cells http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-cells/d3a0fcb/ : SUCCESS in 35m 35s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-full-ceph/b87f77e/ : SUCCESS in 41m 57s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/90/171390/2/check/check-tempest-dsvm-multinode-full/df4efa6/ : FAILURE in 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"48f9acdb011492cc72ad940ffba0f0c36864638b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-04-25 11:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/171390/2 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/171390/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2e4ecf81751b70f07939cb77a3fa7116be67e64b","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-04-26 06:10:49.000000000","message":"Patch Set 2:\n\nI wrote done some analysis in the launchpad: https://bugs.launchpad.net/nova/+bug/1249065/comments/30\n\nI feel the race is between the fixed ip allocation and dhcpbridge script.","accounts_in_message":[],"_revision_number":2},{"id":"b52221d27d19d46f3e1e7a8018f581638858bcba","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-04-26 06:45:20.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI Think that Alex\u0027s anaylysis is correct and am not sure if this actually addresses the problem.","accounts_in_message":[],"_revision_number":2},{"id":"c0b074322596a0ef3893e482a76483ea2c896820","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-27 03:36:20.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/90/171390/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/176a4a217b084669a92a323711277973/index.html : SUCCESS in 21m 51s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/90/171390/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ff4c8b88ca504a5b884470c3b24f9958/index.html : SUCCESS in 19m 05s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/90/171390/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/00bc5d42405d49cb8c32caf1f2062a11/index.html : SUCCESS in 21m 18s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/90/171390/2/check/real-db-upgrade_nova_percona_user_001:th-percona/58c235f42c9d4be4937e099c31f5ad05/index.html : SUCCESS in 22m 12s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/90/171390/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/03cfaec976b44f43999e40f89aeeb504/index.html : SUCCESS in 16m 17s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/90/171390/2/check/real-db-upgrade_nova_percona_user_002:th-percona/6f165ae6bddc46628339287080b70d0c/index.html : SUCCESS in 19m 34s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"065b22b03a199b487df0ce904bda80f742a5a2e4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-27 13:31:12.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ab36b14e01cae01ecfdfd5528ec65513567ae4e0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-28 17:48:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb053f644b897595ce7a2470ed729a9850c30184","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-28 17:52:58.000000000","message":"Patch Set 2: Workflow-1\n\nLooking at Alex\u0027s notes.","accounts_in_message":[],"_revision_number":2},{"id":"1760c05921d341a8b57bfaac923fd8ec6c88ea63","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-04-28 18:08:35.000000000","message":"Patch Set 2:\n\nAlex, I took a look at your notes and that might explain the nova-network case, but it won\u0027t explain the neutron case, which we\u0027re also failing here:\n\nhttp://logs.openstack.org/14/173114/5/check/check-tempest-dsvm-neutron-full/e903e8f/logs/screen-n-api.txt.gz#_2015-04-21_00_40_59_830\n\nIn the case of neutron, we\u0027ve had known races with the neutronv2 API code and the network info cache so refreshing (really rebuilding from neutron) is what would happen here.","accounts_in_message":[],"_revision_number":2},{"id":"546dd91b4b4e3350d0aea967f32e1b29b6bcd23f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-06-16 16:31:00.000000000","message":"Abandoned\n\nThe instance.info_cache is not None, but the network_info attribute inside it is empty:\nhttp://logs.openstack.org/69/190169/3/check/check-tempest-dsvm-full-glusterfs-nv/eecdc4e/logs/screen-n-api.txt.gz#_2015-06-11_14_40_45_576\nThe instance.info_cache is not None, but it\u0027s network_info attribute is empty:\n2015-06-11 14:40:45.576 WARNING nova.api.openstack.compute.contrib.floating_ips [req-b21b23d2-fc5b-4520-84e8-596f994f7873 ServerRescueTestJSON-360939495 ServerRescueTestJSON-1539944143] [instance: a12ee0d4-5747-4e40-9d22-d956afcbe490] Info cache is InstanceInfoCache(created_at\u003d2015-06-11T14:39:49Z,deleted\u003dFalse,deleted_at\u003dNone,instance_uuid\u003da12ee0d4-5747-4e40-9d22-d956afcbe490,network_info\u003dNetworkModel(),updated_at\u003dNone) during associate","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d124c023c16629d08bac180f5f00b75a3349491a","revisions":{"7f5003a817a2ab95424bdb7a929959c6090ff895":{"kind":"REWORK","_number":1,"created":"2015-04-07 21:29:45.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/90/171390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/171390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/171390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/171390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/171390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/171390/1"}}},"commit":{"parents":[{"commit":"bf70df295b0529da9f4381f52fc94d328fa2fdb3","subject":"Merge \"neutronv2: only create client once when adding/removing fixed IPs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf70df295b0529da9f4381f52fc94d328fa2fdb3"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-04-07 21:05:32.000000000","tz":-420},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-04-07 21:05:32.000000000","tz":-420},"subject":"Allow refreshing network info_cache in get_nw_info_for_instance","message":"Allow refreshing network info_cache in get_nw_info_for_instance\n\nThere has been a long-standing race bug in the gate where adding a\nfloating IP to an instance fails with a 400 because instance.info_cache\nis None.\n\nThis change allows a caller to tell get_nw_info_for_instance to get the\nlatest network information from the network API if instance.info_cache\nis not set.\n\nA dependent change to the os-floating-ips extension will use this to\nforce the refresh if the network info_cache is not set on the instance\nrather than error out.\n\nPartial-Bug: #1249065\n\nChange-Id: I326ffaaab941e9fb897a03da3512e4cb481e4a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f5003a817a2ab95424bdb7a929959c6090ff895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f5003a817a2ab95424bdb7a929959c6090ff895"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bf70df295b0529da9f4381f52fc94d328fa2fdb3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d124c023c16629d08bac180f5f00b75a3349491a":{"kind":"REWORK","_number":2,"created":"2015-04-21 21:43:47.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/90/171390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/171390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/171390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/171390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/171390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/171390/2"}}},"commit":{"parents":[{"commit":"2141620d47c919e9cafea9846cabc323142b844e","subject":"Merge \"Fix max_number for migrate_flavor data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2141620d47c919e9cafea9846cabc323142b844e"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-04-07 21:05:32.000000000","tz":-420},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-04-21 20:19:57.000000000","tz":-420},"subject":"Allow refreshing network info_cache in get_nw_info_for_instance","message":"Allow refreshing network info_cache in get_nw_info_for_instance\n\nThere has been a long-standing race bug in the gate where adding a\nfloating IP to an instance fails with a 400 because instance.info_cache\nis None.\n\nThis change allows a caller to tell get_nw_info_for_instance to get the\nlatest network information from the network API if instance.info_cache\nis not set.\n\nA dependent change to the os-floating-ips extension will use this to\nforce the refresh if the network info_cache is not set on the instance\nrather than error out.\n\nPartial-Bug: #1249065\n\nChange-Id: I326ffaaab941e9fb897a03da3512e4cb481e4a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d124c023c16629d08bac180f5f00b75a3349491a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d124c023c16629d08bac180f5f00b75a3349491a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2141620d47c919e9cafea9846cabc323142b844e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
