)]}'
{"id":"openstack%2Fnova~173794","triplet_id":"openstack%2Fnova~master~I4a154e0cae3b8e22bd59ed05ba708e07eed8dea7","project":"openstack/nova","branch":"master","topic":"fira-migration","hashtags":[],"change_id":"I4a154e0cae3b8e22bd59ed05ba708e07eed8dea7","subject":"Resource tracker: unable to restart nova compute","status":"MERGED","created":"2015-04-15 12:18:04.000000000","updated":"2015-04-20 11:03:37.000000000","submitted":"2015-04-20 11:03:34.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed1c24873b6168e181ca597ad7399b0b47e2c686","_number":173794,"virtual_id_number":173794,"owner":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2015-04-20 11:03:34.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-04-16 16:19:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-16 14:46:14.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2015-04-16 13:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":1,"date":"2015-04-16 13:37:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-16 14:11:50.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-16 16:49:16.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-16 13:18:12.000000000","_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":2,"date":"2015-04-20 10:02:20.000000000","_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":1,"date":"2015-04-20 10:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},{"value":2,"date":"2015-04-20 10:05:52.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":1,"date":"2015-04-16 15:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},{"value":1,"date":"2015-04-20 10:05:52.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-04-15 14:35:36.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2015-04-15 18:36:47.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2015-04-15 19:40:06.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2015-04-16 13:37:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-16 13:46:26.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-16 14:11:50.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-16 14:46:14.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-16 15:26:58.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2015-04-16 16:19:13.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-16 16:49:16.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-20 10:02:20.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"},{"updated":"2015-04-20 10:05:52.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2015-04-20 10:24:09.000000000","updated_by":{"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},"reviewer":{"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},"state":"REVIEWER"},{"updated":"2015-04-20 11:03:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"64d9b37809b103225c9f3cf528a188c91f8f6b7e","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-04-15 12:18:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fafc0288e5ccc0cc92111d6e53ed9340c50c07c5","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-04-15 12:21:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI have a feeling the object translation only half happened here, maybe these other bits need tidying up too. Is there a quick unit test we could add to prove this all out?","accounts_in_message":[],"_revision_number":1},{"id":"f72f6f9c29ffccb341836b9ee95e6eeee573e3a5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-15 12:30:16.000000000","message":"Patch Set 1:\n\n* pci-test http://198.175.100.33/173794/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6f90f4642c959d0dfdb45c01b7d176c72064f6da","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-15 12:46:00.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5c540266beb449ebac84120c5e1f690b/index.html : SUCCESS in 22m 29s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b30555d455e4493ea194da2e3170a538/index.html : SUCCESS in 21m 10s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f643a22257f34eaabef58147c6227d8f/index.html : SUCCESS in 22m 39s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/1/check/real-db-upgrade_nova_percona_user_001:th-percona/4b77ceb7af57425ea125c456c61b31f0/index.html : SUCCESS in 27m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3a283424592343778ec11bda9aeaa1e6/index.html : SUCCESS in 19m 35s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/1/check/real-db-upgrade_nova_percona_user_002:th-percona/aac8843a70524054be295ad388a38e26/index.html : SUCCESS in 23m 14s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"ca236380c311a8bc628ec48108862191655f997e","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-04-15 12:47:51.000000000","message":"Patch Set 1:\n\n(2 comments)\n\ni am currently working on unit tests","accounts_in_message":[],"_revision_number":1},{"id":"285d3fa3afdc23a8a70573b68eaa831146aa8071","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-04-15 12:59:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e490c9c1f54c273b44d95972b73408bdd1a81c2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-04-15 13:10:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/173794/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/173794/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f032d99bcaec490cbf7c33042dbeceffea631fec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-15 13:14:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/94/173794/1/check/gate-nova-pep8/bd7a91f/ : SUCCESS in 10m 14s\n- gate-nova-docs http://docs-draft.openstack.org/94/173794/1/check/gate-nova-docs/1815bf1//doc/build/html/ : SUCCESS in 10m 47s\n- gate-nova-python27 http://logs.openstack.org/94/173794/1/check/gate-nova-python27/ce15c80/ : SUCCESS in 17m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-full/c2e9d6a/ : SUCCESS in 43m 09s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-postgres-full/ca4d395/ : SUCCESS in 43m 35s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-neutron-full/412b449/ : SUCCESS in 48m 14s\n- check-grenade-dsvm http://logs.openstack.org/94/173794/1/check/check-grenade-dsvm/2b3993c/ : SUCCESS in 46m 39s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/94/173794/1/check/gate-tempest-dsvm-large-ops/78d1a13/ : SUCCESS in 21m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/94/173794/1/check/gate-tempest-dsvm-neutron-large-ops/d479b93/ : SUCCESS in 21m 46s\n- check-devstack-dsvm-cells http://logs.openstack.org/94/173794/1/check/check-devstack-dsvm-cells/2592370/ : SUCCESS in 18m 25s\n- gate-nova-tox-functional http://logs.openstack.org/94/173794/1/check/gate-nova-tox-functional/d216077/ : SUCCESS in 14m 41s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/94/173794/1/check/check-grenade-dsvm-partial-ncpu/aa54a76/ : SUCCESS in 44m 14s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-ironic-pxe_ssh/05f5eea/ : SUCCESS in 32m 07s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-nova-v21-full/90ad758/ : SUCCESS in 39m 55s\n- check-tempest-dsvm-cells http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-cells/96736e0/ : FAILURE in 43m 25s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-full-ceph/9df1d3f/ : SUCCESS in 35m 47s\n- check-tempest-dsvm-aiopcpu-full http://logs.openstack.org/94/173794/1/check/check-tempest-dsvm-aiopcpu-full/d430609/ : SUCCESS in 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b5d395065923c18e4f6535fb0c04fede8473333","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-04-15 13:40:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/173794/1/ : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dcd314d719c43d98049eee87759e2c4bbfc4394d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-15 13:43:00.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/173794/1/check-ibm-tempest-dsvm-full/0d59300 : SUCCESS in 50m 47s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/173794/1/check-ibm-tempest-dsvm-postgres-full/4fbbdd1 : SUCCESS in 51m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6487ee8b77a3a1464d972d0ebab11ab3afaacb5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2015-04-15 18:36:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThis needs a unit test, IMHO.","accounts_in_message":[],"_revision_number":1},{"id":"acdef2a5df5afc0868b7ddc65ec510104c1f6fde","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-04-16 12:42:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f87e211b9c981b4efdef846b8d4473d5d3effa74","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-16 13:11:15.000000000","message":"Patch Set 2: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/03720175083d4b04ab104fe701489ec3/index.html : SUCCESS in 22m 57s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/eba62aad81a34294aa50b700d418730e/index.html : FAILURE running the job\nException: Failed to fetch patchset in 6m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/30412e3fa50f46f18e7e62e50cc69e5b/index.html : SUCCESS in 23m 42s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_percona_user_001:th-percona/69611ec4c93d486e9d5a95314b00dd0c/index.html : FAILURE running the job\nException: Failed to fetch patchset in 6m 07s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a3d90ea103474ff2ac5f70b1695a4c82/index.html : SUCCESS in 19m 47s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_percona_user_002:th-percona/2ddc2883d4274efbad3199772f1e5a24/index.html : SUCCESS in 23m 19s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"f910104f3c73e2c20dfe43ff2a1f26363cc33743","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-04-16 13:18:12.000000000","message":"Patch Set 2:\n\nrecheck migrations","accounts_in_message":[],"_revision_number":2},{"id":"de2b019f4a460a24a4e95e04207fc9ca8c4aafb7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-04-16 13:37:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/173794/2 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/173794/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0d7fbbbb1fc224ca00b73e16846bf12ae539d3ec","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-16 13:46:26.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/628ba9588ec74e95b12fff0ae3f1cb99/index.html : SUCCESS in 19m 21s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a7461da53782467f92a92453f1882cd5/index.html : SUCCESS in 21m 26s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7612b7d99e844fdd9c3d5060e726d710/index.html : SUCCESS in 23m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_percona_user_001:th-percona/76e3ceb5918644ffa5ff64173bf497d1/index.html : SUCCESS in 27m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/64b98cfeeee24e22bfab22f0dbf2cbaf/index.html : SUCCESS in 23m 32s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/94/173794/2/check/real-db-upgrade_nova_percona_user_002:th-percona/f24b17abb81441d4b9a994fbe76df9cd/index.html : SUCCESS in 22m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c6fe33e1eebc4e6f737fc2f8153da1735f283f2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-04-16 14:11:50.000000000","message":"Patch Set 2:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/94/173794/2/15085/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"abd48af8b903eede201e710a994c58595fedb4ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-16 14:38:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/94/173794/2/check/gate-nova-pep8/f5402e9/ : SUCCESS in 7m 45s\n- gate-nova-docs http://docs-draft.openstack.org/94/173794/2/check/gate-nova-docs/8e36856//doc/build/html/ : SUCCESS in 8m 25s\n- gate-nova-python27 http://logs.openstack.org/94/173794/2/check/gate-nova-python27/5ca0c62/ : SUCCESS in 15m 15s\n- check-tempest-dsvm-full http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-full/cb07300/ : SUCCESS in 48m 57s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-postgres-full/5a585c1/ : SUCCESS in 48m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-neutron-full/746ffe5/ : SUCCESS in 1h 02m 23s\n- check-grenade-dsvm http://logs.openstack.org/94/173794/2/check/check-grenade-dsvm/20bbc6f/ : SUCCESS in 43m 07s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/94/173794/2/check/gate-tempest-dsvm-large-ops/38e3f10/ : SUCCESS in 22m 57s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/94/173794/2/check/gate-tempest-dsvm-neutron-large-ops/be9232e/ : SUCCESS in 22m 23s\n- check-devstack-dsvm-cells http://logs.openstack.org/94/173794/2/check/check-devstack-dsvm-cells/688df1a/ : SUCCESS in 19m 01s\n- gate-nova-tox-functional http://logs.openstack.org/94/173794/2/check/gate-nova-tox-functional/ffd472a/ : SUCCESS in 11m 57s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/94/173794/2/check/check-grenade-dsvm-partial-ncpu/6e7f5e4/ : SUCCESS in 38m 48s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-ironic-pxe_ssh/f2df721/ : SUCCESS in 34m 24s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-nova-v21-full/fb17fee/ : SUCCESS in 53m 10s\n- check-tempest-dsvm-cells http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-cells/ec1ae52/ : FAILURE in 43m 09s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-full-ceph/b5d7e5c/ : SUCCESS in 39m 26s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/94/173794/2/check/check-tempest-dsvm-multinode-full/11fd65e/ : SUCCESS in 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"839566742ee4accaf92cd04216a2ed37f201d3da","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-16 14:46:14.000000000","message":"Patch Set 2:\n\n* pci-test http://198.175.100.33/173794/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"dd7067bda73ca364fa98def64114ae8871e3f5aa","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-04-16 15:26:58.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"ba1b00fd063f17cfdb46ac8d2fe9e99097912f1d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-16 16:19:13.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/173794/2/check-ibm-tempest-dsvm-full/908560c : SUCCESS in 1h 07m 19s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/173794/2/check-ibm-tempest-dsvm-postgres-full/7ae2f07 : SUCCESS in 1h 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"810f168d1d283b0a1ab97a38be080ed4e2dabe87","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-04-16 16:49:16.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/173794/2/ : FAILURE in 2h 23m 50s","accounts_in_message":[],"_revision_number":2},{"id":"81590e8e5885389b96e2a1661e47bb02d8b4057e","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2015-04-20 08:47:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bed71fc24600a5b857ba239c66a803daf291e6e","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2015-04-20 10:02:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73ce284ac47cf5680b92b37839a483b1ff2b33dd","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-04-20 10:05:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)\n\nI have a nit in the unit test, but its an important fix, lets try get this merged.","accounts_in_message":[],"_revision_number":2},{"id":"3b2d0c6702c7223660fb3f54528ac4b72657af07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-20 10:05:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"99157b9056fa01fc75e6f5777c8b439473ca61af","author":{"_account_id":10795,"name":"Rajesh Tailor","email":"tailor.rajesh88@gmail.com","username":"rajeshtailor"},"date":"2015-04-20 10:24:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aaaaa8b6ace1f352511797a1d19295d6aba7dab1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-20 11:03:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/94/173794/2/gate/gate-nova-docs/27ba0e9//doc/build/html/ : SUCCESS in 8m 04s\n- gate-nova-pep8 http://logs.openstack.org/94/173794/2/gate/gate-nova-pep8/98ebe21/ : SUCCESS in 8m 05s\n- gate-nova-python27 http://logs.openstack.org/94/173794/2/gate/gate-nova-python27/8f122b6/ : SUCCESS in 12m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-full/f9540bd/ : SUCCESS in 46m 28s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-postgres-full/87031a3/ : SUCCESS in 46m 50s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-neutron-full/59e03a0/ : SUCCESS in 55m 20s\n- gate-grenade-dsvm http://logs.openstack.org/94/173794/2/gate/gate-grenade-dsvm/4cdb836/ : SUCCESS in 36m 02s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-large-ops/05edf75/ : SUCCESS in 21m 44s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-neutron-large-ops/54ab380/ : SUCCESS in 20m 24s\n- gate-nova-tox-functional http://logs.openstack.org/94/173794/2/gate/gate-nova-tox-functional/0af9a4a/ : SUCCESS in 12m 05s\n- gate-devstack-dsvm-cells http://logs.openstack.org/94/173794/2/gate/gate-devstack-dsvm-cells/14e2bdf/ : SUCCESS in 22m 08s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/94/173794/2/gate/gate-grenade-dsvm-partial-ncpu/01e53dc/ : SUCCESS in 37m 37s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-nova-v21-full/d07676a/ : SUCCESS in 47m 02s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/94/173794/2/gate/gate-tempest-dsvm-full-ceph/b53a457/ : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a42da774bf4b06b7ae8c2b3c89113879cc1e574c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-20 11:03:37.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee7a7446cc6947a6bacacb6cb514934cc22e5782","revisions":{"26fe91a2ef11b81956bd5a8bd48546b883c0e08f":{"kind":"REWORK","_number":1,"created":"2015-04-15 12:18:04.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/94/173794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/173794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/173794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/173794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/173794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/173794/1"}}},"commit":{"parents":[{"commit":"feef80c1f88cc28f8625b19ebf43b190fae0ca08","subject":"Merge \"Forbid booting of QCOW2 images with virtual_size \u003e root_gb\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/feef80c1f88cc28f8625b19ebf43b190fae0ca08"}]}],"author":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2015-04-15 12:14:42.000000000","tz":-420},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2015-04-15 12:17:41.000000000","tz":-420},"subject":"Resource tracker: unable to restart nova compute","message":"Resource tracker: unable to restart nova compute\n\nThe resource tracker calculates its used reousrces. In certain cases\nof failed migrations and an instance being deleted the resource tracker\ncauses an exception in nova compute. If this situation arises then nova\ncompute may not even be able to restart.\n\nChange-Id: I4a154e0cae3b8e22bd59ed05ba708e07eed8dea7\nCloses-bug: #1444439\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26fe91a2ef11b81956bd5a8bd48546b883c0e08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26fe91a2ef11b81956bd5a8bd48546b883c0e08f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"feef80c1f88cc28f8625b19ebf43b190fae0ca08","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ee7a7446cc6947a6bacacb6cb514934cc22e5782":{"kind":"REWORK","_number":2,"created":"2015-04-16 12:42:05.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/94/173794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/173794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/173794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/173794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/173794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/173794/2"}}},"commit":{"parents":[{"commit":"d1fb8d0fbdd6cb95c43b02f754409f1c728e8cd0","subject":"compute: stop handling virt lifecycle events in cleanup_host()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1fb8d0fbdd6cb95c43b02f754409f1c728e8cd0"}]}],"author":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2015-04-15 12:14:42.000000000","tz":-420},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2015-04-16 12:41:27.000000000","tz":-420},"subject":"Resource tracker: unable to restart nova compute","message":"Resource tracker: unable to restart nova compute\n\nThe resource tracker calculates its used resources. In certain cases\nof failed migrations and an instance being deleted the resource tracker\ncauses an exception in nova compute. If this situation arises then nova\ncompute may not even be able to restart.\n\nChange-Id: I4a154e0cae3b8e22bd59ed05ba708e07eed8dea7\nCloses-bug: #1444439\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee7a7446cc6947a6bacacb6cb514934cc22e5782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee7a7446cc6947a6bacacb6cb514934cc22e5782"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d1fb8d0fbdd6cb95c43b02f754409f1c728e8cd0","is_merged_in_target_branch":true,"change_id":"I1a321371dff7933cdd11d31d9f9c2a2f850fd8d9","change_number":174069,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
