)]}'
{"id":"openstack%2Fnova~175780","triplet_id":"openstack%2Fnova~master~Id54f7bdfa14a19da41da554b13ba9496ee525c71","project":"openstack/nova","branch":"master","topic":"bug/1298242","hashtags":[],"change_id":"Id54f7bdfa14a19da41da554b13ba9496ee525c71","subject":"libvirt: make live_migration_uri flag dependent on virt_type","status":"MERGED","created":"2015-04-21 09:05:11.000000000","updated":"2016-02-19 12:36:05.000000000","submitted":"2016-02-18 15:09:17.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9522d5f08848524417a61b70d4221bc30948d311","_number":175780,"virtual_id_number":175780,"owner":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-02-18 15:09:17.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":1,"date":"2016-02-15 10:04:06.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"date":"2016-02-19 12:36:05.000000000","post_submit":true,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2016-02-15 09:07:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"all":[{"value":0,"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2016-02-15 11:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":2,"date":"2016-02-18 14:09:23.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-02-15 11:48:57.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":1,"date":"2016-02-18 14:09:23.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-06-04 15:32:50.000000000","updated_by":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"reviewer":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"state":"REVIEWER"},{"updated":"2015-08-04 12:09:21.000000000","updated_by":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"reviewer":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"state":"REVIEWER"},{"updated":"2015-08-28 06:19:08.000000000","updated_by":{"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},"reviewer":{"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},"state":"REVIEWER"},{"updated":"2015-09-04 08:53:39.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2015-09-06 12:00:29.000000000","updated_by":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"reviewer":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"state":"REVIEWER"},{"updated":"2015-09-22 02:26:07.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2015-09-29 19:44:33.000000000","updated_by":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"reviewer":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"state":"REVIEWER"},{"updated":"2015-10-06 16:02:17.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2015-11-16 12:53:34.000000000","updated_by":{"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},"reviewer":{"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},"state":"REVIEWER"},{"updated":"2015-12-15 17:54:08.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-12-23 10:08:25.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-19 10:57:42.000000000","updated_by":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"reviewer":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"state":"REVIEWER"},{"updated":"2016-01-27 13:21:54.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-28 09:06:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-07 14:32:25.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-11 12:35:58.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-12 04:36:54.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-12 10:22:25.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-12 11:08:57.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-12 11:43:46.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-12 20:57:30.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-15 09:07:21.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-15 10:04:06.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-15 11:48:57.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2016-02-15 11:50:01.000000000","updated_by":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"reviewer":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"state":"REVIEWER"},{"updated":"2016-02-18 14:09:23.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-02-18 15:09:17.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"eb93b97bbbb6b154c63b87f1108fcbf2a53e4ec1","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-04-21 09:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b483f4011cc468aac3e7a5d93589449d506c848b","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-04-21 09:16:44.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5072a95e41e29e4e66eca1ac2371dde09125416a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-04-21 09:37:26.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a32a4ec26c2c4ea18df0a526ef2a27c2/index.html : SUCCESS in 13m 49s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/159f9a9866394919a61e57292b80b3d1/index.html : SUCCESS in 16m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e5289ceac23847dd8c5021ee38e2d96a/index.html : SUCCESS in 16m 27s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/1/check/real-db-upgrade_nova_percona_user_001:th-percona/bedb58a228574c65adc875862cf3e3c6/index.html : SUCCESS in 15m 42s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/97e2ed38b89b4229ad957c90e51e5cd3/index.html : SUCCESS in 14m 19s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/1/check/real-db-upgrade_nova_percona_user_002:th-percona/77e9557138444069982cc5f2258d5e6c/index.html : SUCCESS in 14m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"7a673f5a1488d17025dfe9750821de04f87baf69","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-04-21 09:54:21.000000000","message":"Patch Set 1:\n\n* pci-test http://198.175.100.33/175780/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"768a19d77680f9b20a53db36fc7e9d1ee4e38961","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-21 10:05:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/1/check/gate-nova-pep8/24944b4/ : SUCCESS in 11m 16s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/1/check/gate-nova-docs/be60d3f//doc/build/html/ : SUCCESS in 9m 35s\n- gate-nova-python27 http://logs.openstack.org/80/175780/1/check/gate-nova-python27/cd132e5/ : FAILURE in 14m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-full/c861778/ : SUCCESS in 44m 01s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-postgres-full/837e3dd/ : SUCCESS in 42m 04s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-neutron-full/dddefeb/ : SUCCESS in 57m 26s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/1/check/check-grenade-dsvm/8719c8b/ : SUCCESS in 42m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/1/check/gate-tempest-dsvm-large-ops/d75999f/ : SUCCESS in 26m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/1/check/gate-tempest-dsvm-neutron-large-ops/7df653c/ : SUCCESS in 24m 17s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/1/check/check-devstack-dsvm-cells/8b6c14e/ : SUCCESS in 20m 22s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/1/check/gate-nova-tox-functional/d555135/ : SUCCESS in 12m 34s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/1/check/check-grenade-dsvm-partial-ncpu/60c6880/ : SUCCESS in 39m 10s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-ironic-pxe_ssh/9e6bab4/ : SUCCESS in 36m 02s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-nova-v21-full/f650712/ : SUCCESS in 42m 37s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-cells/7d82fdb/ : FAILURE in 41m 33s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-full-ceph/eb4d9d7/ : SUCCESS in 39m 21s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/1/check/check-tempest-dsvm-multinode-full/5daf1c9/ : FAILURE in 51m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e93ae27951903fb0ae973c8b57bcd427650890b2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-21 10:05:47.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/1/check-ibm-tempest-dsvm-full/4b8138f : FAILURE in 11s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/1/check-ibm-tempest-dsvm-postgres-full/4433a93 : SUCCESS in 1h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34df651b7a6884d98869a074b0b8cc15c7765c6c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-04-21 10:11:04.000000000","message":"Patch Set 1:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/1/15389/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"1dd66bcde8c88bb07d5ad1216d868ba2cc815e5f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-04-21 11:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/1/passonly/dsvm-tempest-xen/8f2644b : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d9ea45274e72da4d2e662e4f4e3777d9ef086113","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-04-21 11:24:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/175780/1/ : FAILURE in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"defdbeab396120a4d3faf9f6100c3cc1003b12d3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-04-24 17:59:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1422b77f84f74e97f5a7bb7501c72ade73e14662","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-04 10:30:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc8eaacf0e33aeb07c3b811b02bc21aec32558d8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-04 10:33:24.000000000","message":"Patch Set 2:\n\n* pci-test http://198.175.100.33/175780/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1724647c378c986c0ce5ee238dc6fba7d133b65e","author":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"date":"2015-06-04 11:09:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nComment inline. Also, spelling nit: You want \"dependEnt\" (\"dependAnt\" is a different word and might confuse translation efforts as these are user-facing help strings.)","accounts_in_message":[],"_revision_number":2},{"id":"fd8e21923ab856860968d09bd44c00f26e196f46","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-04 11:19:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d807273bf5d8261168a0829de549fb2a781e89b","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-04 11:21:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ff0fcec80fc4049621ce85fb499317fc1e143f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-04 11:52:28.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/3/check-ibm-tempest-dsvm-full/5a29cde : FAILURE in 28m 43s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/3/check-ibm-tempest-dsvm-postgres-full/ea5d583 : FAILURE in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ee408e6f4859e261a66447f012122919b9ea347","author":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"date":"2015-06-04 12:27:48.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks for your explanation there Alvaro - I had guessed that might be the case. I agree that sticking with the current default for non-supported drivers is as sensible as any other choice.","accounts_in_message":[],"_revision_number":3},{"id":"ee4d843a1e3ef7000ec1ae2e1e0ad497baa6a20d","author":{"_account_id":8276,"name":"kiwik","email":"chenrui.momo@gmail.com","username":"kiwik"},"date":"2015-06-04 12:28:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nA typo issue, and if you can add test case to cover the new function, I think it make sense.","accounts_in_message":[],"_revision_number":3},{"id":"9a361f3a03d9f22f4fd2e5615cdad9b0cc937cfe","author":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"date":"2015-06-04 12:40:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b781745c4da3e464e24d2b7aa66485bcfd76e677","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-04 12:46:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/3/check/gate-nova-pep8/a0b0a73/ : SUCCESS in 11m 00s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/3/check/gate-nova-docs/102ddb3//doc/build/html/ : SUCCESS in 12m 02s\n- gate-nova-python27 http://logs.openstack.org/80/175780/3/check/gate-nova-python27/bb7784d/ : SUCCESS in 15m 53s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-full/da2f19d/ : SUCCESS in 48m 01s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-postgres-full/fb18f1e/ : SUCCESS in 1h 04m 29s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-neutron-full/c30057a/ : SUCCESS in 43m 05s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/3/check/check-grenade-dsvm/7cfc8bc/ : SUCCESS in 37m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/3/check/gate-tempest-dsvm-large-ops/926fd6c/ : SUCCESS in 28m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/3/check/gate-tempest-dsvm-neutron-large-ops/c46206d/ : SUCCESS in 23m 50s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/3/check/check-devstack-dsvm-cells/c99cae8/ : SUCCESS in 42m 00s\n- gate-nova-python34 http://logs.openstack.org/80/175780/3/check/gate-nova-python34/3b833d6/ : SUCCESS in 9m 24s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/3/check/gate-nova-tox-functional/dc117c2/ : SUCCESS in 9m 46s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/3/check/check-grenade-dsvm-partial-ncpu/bdbeddf/ : SUCCESS in 36m 53s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-ironic-pxe_ssh/3b6eb91/ : SUCCESS in 32m 55s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-nova-v21-full/a6b429a/ : SUCCESS in 1h 07m 12s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-cells/0776f84/ : SUCCESS in 36m 59s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-full-ceph/822f01f/ : SUCCESS in 33m 59s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/3/check/check-tempest-dsvm-multinode-full/3ca28a3/ : SUCCESS in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d206ced9220e57216b72453e2abd7118e7adee71","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-04 13:03:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/3/check/dsvm-tempest-xen/639645e : SUCCESS in 1h 30m 59s","accounts_in_message":[],"_revision_number":3},{"id":"4b6518afb68b07e3d9f8ea7e7c44aaf37eedaf65","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-04 13:09:18.000000000","message":"Patch Set 3:\n\n* pci-test http://198.175.100.33/175780/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5deb3b895d920a88da613a760f81f236002e7c2c","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-04 13:59:28.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2eda1b9e2597a303df1fb8a821c67f2aa221a5c5","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-04 14:57:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e080f18144a599288faff6e2fdf5d304978b933","author":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"date":"2015-06-04 15:32:50.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5f2e991e3be4f2db69d609b576a2733751b8a47c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-04 15:49:39.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/4/check-ibm-tempest-dsvm-full/59ddcc3 : SUCCESS in 49m 22s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/4/check-ibm-tempest-dsvm-postgres-full/bc3fa17 : SUCCESS in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d942cf50c9493d2a7ed2c595809b749e739f599a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-04 16:01:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/4/check/gate-nova-pep8/d4df2ac/ : SUCCESS in 9m 04s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/4/check/gate-nova-docs/b3d1930//doc/build/html/ : SUCCESS in 8m 43s\n- gate-nova-python27 http://logs.openstack.org/80/175780/4/check/gate-nova-python27/bbd1197/ : SUCCESS in 13m 49s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-full/7f550f3/ : SUCCESS in 49m 23s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-postgres-full/97715e2/ : SUCCESS in 43m 57s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-neutron-full/1eccba3/ : SUCCESS in 56m 50s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/4/check/check-grenade-dsvm/9193f14/ : SUCCESS in 38m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/4/check/gate-tempest-dsvm-large-ops/ef25bec/ : SUCCESS in 37m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/4/check/gate-tempest-dsvm-neutron-large-ops/6aaeea7/ : SUCCESS in 28m 07s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/4/check/check-devstack-dsvm-cells/61d41a9/ : SUCCESS in 18m 31s\n- gate-nova-python34 http://logs.openstack.org/80/175780/4/check/gate-nova-python34/45087c6/ : SUCCESS in 8m 59s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/4/check/gate-nova-tox-functional/e7374c8/ : SUCCESS in 12m 05s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/4/check/check-grenade-dsvm-partial-ncpu/8d357a2/ : SUCCESS in 36m 53s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-ironic-pxe_ssh/c4b7d89/ : SUCCESS in 44m 40s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-nova-v21-full/a8bb898/ : SUCCESS in 41m 19s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-cells/2e28904/ : SUCCESS in 50m 11s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-full-ceph/46a0f0d/ : SUCCESS in 33m 43s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/4/check/check-tempest-dsvm-multinode-full/ffeb5dd/ : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4812160a59fedfda2449fe8e47e6a030eb1f4f8d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-04 16:14:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/4/check/dsvm-tempest-xen/a441f8a : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9caf2f39e54594d0563e607ba7169ba295efb96f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-04 17:18:45.000000000","message":"Patch Set 4:\n\n* pci-test http://198.175.100.33/175780/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ca1bacbc0dbe5d5c61106a5bfbaa7a7462535a9e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-04 19:21:41.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/57cae7ba171f4fc68c88f04356639277/index.html : SUCCESS in 17m 10s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/19098f24a00c42cf8f8c42375d2798bf/index.html : SUCCESS in 26m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/43e26dfc90f04c31bf587b61d4dc332b/index.html : SUCCESS in 19m 54s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/4/check/real-db-upgrade_nova_percona_user_001:th-percona/021ed2328d134f9e91a9555b6865760d/index.html : SUCCESS in 25m 21s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/04cbd587473846d9ac6c67d23765ffdd/index.html : SUCCESS in 16m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/4/check/real-db-upgrade_nova_percona_user_002:th-percona/3dce04a41ca64062a5a69609b0b2c429/index.html : SUCCESS in 16m 52s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"d23f82d9a25cc0564b7e462967836ab301c39768","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-04 21:44:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/175780/4/ : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":4},{"id":"5e77f6ee396496f6203a0078e711b462e3029ffc","author":{"_account_id":8276,"name":"kiwik","email":"chenrui.momo@gmail.com","username":"kiwik"},"date":"2015-06-05 03:09:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1d448047e3f74043d27d9045ccb13869b7381a67","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-11 08:38:01.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"14f806cb22d5fb088e3becddae9e8d49a33bea71","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-06-11 16:32:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/4 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"f9553505920f27137b0817acebe0c2df2122c87f","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-19 08:50:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"538f7b2341c277c1f6c7447fbb7c552661282ebc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-19 09:20:59.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/953ce5f3b695400bbc1f225cb40036a6/index.html : SUCCESS in 20m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0fc25a4c1003416daa1524cd44aaa057/index.html : SUCCESS in 27m 38s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2772f2ca528d40359e901dc4a740434d/index.html : SUCCESS in 21m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/5/check/real-db-upgrade_nova_percona_user_001:th-percona/969571008a354cf2ae349640473df9e8/index.html : SUCCESS in 29m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/559d7e5387204609b802c82bba46073a/index.html : SUCCESS in 18m 35s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/5/check/real-db-upgrade_nova_percona_user_002:th-percona/b5070eaf0d474b4982461251f200e900/index.html : SUCCESS in 22m 30s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"ea1a3684742ee8a7978e44a60162c3aa42630faa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-06-19 09:33:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/5/Nova-ML2-Sriov : SUCCESS in 22m 19s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"9b45afda1b7b52fda13173074ae797643a251961","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-19 09:40:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/5/check/gate-nova-pep8/8e89f48/ : SUCCESS in 6m 22s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/5/check/gate-nova-docs/3d44b39//doc/build/html/ : SUCCESS in 6m 38s\n- gate-nova-python27 http://logs.openstack.org/80/175780/5/check/gate-nova-python27/6362d4a/ : SUCCESS in 10m 33s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-full/61b0f91/ : SUCCESS in 37m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-postgres-full/2c98473/ : SUCCESS in 34m 24s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-neutron-full/6fa5dbf/ : SUCCESS in 45m 19s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/5/check/check-grenade-dsvm/9ecb4b5/ : SUCCESS in 48m 45s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/5/check/gate-tempest-dsvm-large-ops/d301c7d/ : SUCCESS in 20m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/5/check/gate-tempest-dsvm-neutron-large-ops/78a2e7d/ : SUCCESS in 19m 59s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/5/check/check-devstack-dsvm-cells/99921a6/ : SUCCESS in 16m 18s\n- gate-nova-python34 http://logs.openstack.org/80/175780/5/check/gate-nova-python34/ea80e06/ : SUCCESS in 9m 52s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/5/check/gate-nova-tox-functional/9882563/ : SUCCESS in 12m 11s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/5/check/check-grenade-dsvm-partial-ncpu/cf3fb6d/ : SUCCESS in 37m 00s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-ironic-pxe_ssh/a9f2893/ : SUCCESS in 34m 46s\n- check-tempest-dsvm-docker http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-docker/c14a6e3/ : SUCCESS in 33m 44s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-nova-v21-full/67d6522/ : SUCCESS in 43m 56s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-cells/ab78432/ : SUCCESS in 36m 05s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-full-ceph/d9fcd3d/ : SUCCESS in 45m 37s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/5/check/check-tempest-dsvm-multinode-full/c57b992/ : SUCCESS in 42m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d80efba656f4be8db1f7513aebe042fcce1945ee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-19 09:45:52.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/5/check-ibm-tempest-dsvm-full/1efef67 : SUCCESS in 54m 00s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/5/check-ibm-tempest-dsvm-postgres-full/cc5733a : FAILURE in 54m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f3f562b6a3e6565d9fe7b3ab1af9620412bcbfe4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-19 09:49:03.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/5/1717/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"57c6427846c5cfce405507dade1bfe74a9f40cdc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-19 10:06:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/5/check/dsvm-tempest-xen/cd48aeb : FAILURE in 1h 15m 08s","accounts_in_message":[],"_revision_number":5},{"id":"663e93d0255cb12bd23382a55305960b2e267e2c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-06-19 11:45:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/80/175780/5/check-upstream/dsvm-tempest-cinder/49aa697/ : SUCCESS in 1h 07m 21s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"7cf8b07c42495650bbf0c505ff8c4918767b5687","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-19 12:08:30.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/175780/5/ : FAILURE in 2h 36m 42s","accounts_in_message":[],"_revision_number":5},{"id":"ba07b22199ece36f2b18c986c4c467348b7b2651","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-06-19 15:22:16.000000000","message":"Patch Set 5:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":5},{"id":"37c062ea768146de9c8b91f78327c8af1c66ee9d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-19 16:38:57.000000000","message":"Patch Set 5:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/5/1740/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"cf2400d4d6c43fec45f4bb1994cedcae007bdda7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-06-22 23:16:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/5 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"fd67ba1e1f1b00722ef04fafd778cc7c88e9e7d4","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-29 11:47:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"897f10b816a9b06b562aa2cdc1ed2da0494e0cce","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-29 12:18:28.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/867a3cd8d56e4bfc8a308a3f840aaeab/index.html : SUCCESS in 18m 50s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/50a93b57e37e417e97098cea84e79158/index.html : SUCCESS in 23m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/702811df178d492fa1be3a0109ad7277/index.html : SUCCESS in 20m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/6/check/real-db-upgrade_nova_percona_user_001:th-percona/bedba7630cfa4ac8a15396eb30abc786/index.html : SUCCESS in 23m 14s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3f2f404f14164b04b699ac01fa9351d1/index.html : SUCCESS in 22m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/6/check/real-db-upgrade_nova_percona_user_002:th-percona/82b7235050b94dd1ab572090a0305b2a/index.html : SUCCESS in 18m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"609b2c304c7e9368d4a1dcd6b37b9a6522394744","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-29 12:42:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/6/check/gate-nova-pep8/ba8372e/ : SUCCESS in 8m 45s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/6/check/gate-nova-docs/b4746bd//doc/build/html/ : SUCCESS in 9m 01s\n- gate-nova-python27 http://logs.openstack.org/80/175780/6/check/gate-nova-python27/37db554/ : SUCCESS in 13m 44s\n- gate-nova-python34 http://logs.openstack.org/80/175780/6/check/gate-nova-python34/a4c8a98/ : SUCCESS in 10m 13s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-full/a90e4f2/ : SUCCESS in 50m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-postgres-full/4b8e0ce/ : SUCCESS in 32m 13s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-neutron-full/49d18fe/ : SUCCESS in 42m 52s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/6/check/check-grenade-dsvm/b2b3dba/ : SUCCESS in 36m 07s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/6/check/gate-tempest-dsvm-large-ops/527bd9b/ : SUCCESS in 21m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/6/check/gate-tempest-dsvm-neutron-large-ops/4aa05dc/ : SUCCESS in 20m 26s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/6/check/check-devstack-dsvm-cells/b604b9a/ : SUCCESS in 19m 30s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/6/check/gate-nova-tox-functional/164ab9f/ : SUCCESS in 10m 45s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/6/check/check-grenade-dsvm-partial-ncpu/d6ab0ba/ : FAILURE in 36m 49s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-ironic-pxe_ssh/b8ff794/ : SUCCESS in 35m 37s\n- check-tempest-dsvm-docker http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-docker/2999b5a/ : FAILURE in 32m 14s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-nova-v21-full/30ba6f2/ : SUCCESS in 41m 48s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-cells/878c0aa/ : SUCCESS in 39m 34s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-full-ceph/96afe3b/ : SUCCESS in 41m 28s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-multinode-full/3057aaa/ : SUCCESS in 39m 16s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-multinode-smoke/d8c3955/ : SUCCESS in 26m 36s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-neutron-multinode-smoke/59e450a/ : SUCCESS in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fff588b169365a741bcc1b34b6b69f2293a0be48","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-29 12:42:35.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-partial-ncpu: https://bugs.launchpad.net/bugs/1463631\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"66cf3436135d987f8df9b6d1ff4a73a8b5d2895e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-29 12:54:19.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/6/check-ibm-tempest-dsvm-full/4e97101 : SUCCESS in 1h 04m 48s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/6/check-ibm-tempest-dsvm-postgres-full/51d6d13 : SUCCESS in 1h 00m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"913aab1378c18cda08604b03581a409f5eeefd5c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-29 13:00:29.000000000","message":"Patch Set 6: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/6/2624/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"05f1a0c40799ccf160aa45b052284b1f577315f5","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-29 13:04:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/6/check/dsvm-tempest-xen/daf7f01 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":6},{"id":"b588dc325682816b5e4d2ad42a4eaa363aa621b8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-29 13:38:21.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/175780/6/ : FAILURE in 1h 48m 51s","accounts_in_message":[],"_revision_number":6},{"id":"5770cbd8f5b9d518404b5c48d4a7fc2afde1f22b","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-06-30 11:34:43.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b9b423f7883f10523a72c1886599297c382ff96e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-30 12:07:13.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/175780/6/ : FAILURE in 32m 07s","accounts_in_message":[],"_revision_number":6},{"id":"248a3fd980200333537a5c5adae084163244cdfd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-30 12:28:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/6/check/gate-nova-pep8/dfbbe07/ : SUCCESS in 6m 45s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/6/check/gate-nova-docs/0d9cc34//doc/build/html/ : SUCCESS in 9m 27s\n- gate-nova-python27 http://logs.openstack.org/80/175780/6/check/gate-nova-python27/d239425/ : FAILURE in 13m 46s\n- gate-nova-python34 http://logs.openstack.org/80/175780/6/check/gate-nova-python34/86a3e92/ : UNSTABLE in 9m 41s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-full/d590143/ : UNSTABLE in 50m 05s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-postgres-full/f4266fd/ : UNSTABLE in 43m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-neutron-full/f2241ec/ : UNSTABLE in 43m 24s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/6/check/check-grenade-dsvm/51c95c7/ : UNSTABLE in 47m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/6/check/gate-tempest-dsvm-large-ops/962b627/ : UNSTABLE in 26m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/6/check/gate-tempest-dsvm-neutron-large-ops/bea234d/ : UNSTABLE in 24m 31s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/6/check/check-devstack-dsvm-cells/77f5147/ : UNSTABLE in 17m 28s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/6/check/gate-nova-tox-functional/5a1e178/ : FAILURE in 12m 33s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/6/check/check-grenade-dsvm-partial-ncpu/143c9ef/ : UNSTABLE in 40m 53s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-ironic-pxe_ssh/50749c7/ : UNSTABLE in 44m 44s\n- check-tempest-dsvm-docker http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-docker/3bca0f5/ : UNSTABLE in 39m 34s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-nova-v21-full/da688a2/ : UNSTABLE in 46m 16s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-cells/87a7a20/ : UNSTABLE in 42m 57s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-full-ceph/28fb494/ : UNSTABLE in 43m 56s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-multinode-full/a7166ef/ : UNSTABLE in 40m 28s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-multinode-smoke/6c6e140/ : UNSTABLE in 25m 45s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/80/175780/6/check/check-tempest-dsvm-neutron-multinode-smoke/9323df0/ : UNSTABLE in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7bb9e9a4b1dd0f96ef68c5d3548f1eba8e7acf09","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-30 12:44:45.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27: https://bugs.launchpad.net/bugs/1469865\n- gate-nova-tox-functional: https://bugs.launchpad.net/bugs/1469865\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"5d142ce9db0cfdd41e953d561e107bf242218cb0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-03 21:24:06.000000000","message":"Patch Set 6:\n\nBuild failed. Logs: http://208.91.1.172/logs/175780/6 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"ec55be9acc6bcc7c49be598db8c3eb11cad1c5ba","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-07-06 08:05:37.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4b2e8bdb44ee685f7c7fb3cc693dc19bd26ebba7","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-06 08:37:13.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bacc3b76386a46f3ad70813e21f6f76b/index.html : SUCCESS in 17m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/623fee76eda7494cb564d738da907d49/index.html : SUCCESS in 24m 22s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3e0575be49cd416288b8310ca07ef898/index.html : SUCCESS in 17m 22s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/7/check/real-db-upgrade_nova_percona_user_001:th-percona/41f7a562fc9a433aaf44369482f050f0/index.html : SUCCESS in 24m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d8ad1ece64584d50ade5bd9ea4bb1f92/index.html : SUCCESS in 19m 18s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/7/check/real-db-upgrade_nova_percona_user_002:th-percona/5881cf16a9934ecdbdb3c27786dc8b13/index.html : SUCCESS in 17m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"457faa31c25b79b7118a56fa8802ef078d1d3580","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-06 09:03:12.000000000","message":"Patch Set 7: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/7/3382/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":7},{"id":"ee09c5ee0f437263509d019b0194dc2992a85a92","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-06 09:03:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/7/check/gate-nova-pep8/7a8bb1b/ : SUCCESS in 9m 32s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/7/check/gate-nova-docs/1e04b77//doc/build/html/ : SUCCESS in 9m 58s\n- gate-nova-python27 http://logs.openstack.org/80/175780/7/check/gate-nova-python27/4de3715/ : SUCCESS in 12m 03s\n- gate-nova-python34 http://logs.openstack.org/80/175780/7/check/gate-nova-python34/2fa1073/ : SUCCESS in 8m 32s\n- check-tempest-dsvm-full http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-full/b675b33/ : SUCCESS in 39m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-postgres-full/90b5df4/ : SUCCESS in 42m 57s\n- check-tempest-dsvm-neutron http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-neutron/75b2186/ : SUCCESS in 40m 02s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-neutron-full/1188130/ : SUCCESS in 52m 38s\n- check-grenade-dsvm http://logs.openstack.org/80/175780/7/check/check-grenade-dsvm/49046f7/ : SUCCESS in 46m 27s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/7/check/gate-tempest-dsvm-large-ops/cc9dc6a/ : SUCCESS in 34m 51s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/7/check/gate-tempest-dsvm-neutron-large-ops/411cabe/ : SUCCESS in 30m 54s\n- check-devstack-dsvm-cells http://logs.openstack.org/80/175780/7/check/check-devstack-dsvm-cells/0b02e54/ : SUCCESS in 25m 59s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/7/check/gate-nova-tox-functional/6ac248a/ : SUCCESS in 10m 07s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/7/check/check-grenade-dsvm-partial-ncpu/b597ce5/ : SUCCESS in 53m 13s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-ironic-pxe_ssh/288897b/ : SUCCESS in 44m 24s\n- check-tempest-dsvm-docker http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-docker/180212a/ : SUCCESS in 41m 52s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-nova-v21-full/3017626/ : SUCCESS in 53m 02s\n- check-tempest-dsvm-cells http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-cells/c2c579a/ : SUCCESS in 35m 02s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-full-ceph/353473d/ : SUCCESS in 50m 50s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-multinode-full/fae618a/ : FAILURE in 22m 22s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-multinode-smoke/2dea7be/ : FAILURE in 20m 47s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/80/175780/7/check/check-tempest-dsvm-neutron-multinode-smoke/6548b28/ : FAILURE in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5910a3ce7797eaa8ac4359f8a49a1add619f6ca6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-06 09:16:41.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/7/check-ibm-tempest-dsvm-full/d11d3c7 : SUCCESS in 53m 42s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/7/check-ibm-tempest-dsvm-postgres-full/7d8b7a5 : SUCCESS in 1h 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f4df6e92b565f67537537615c4007e590c6a6695","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-06 09:25:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/7/check/dsvm-tempest-xen/7f76ded : FAILURE in 1h 19m 17s","accounts_in_message":[],"_revision_number":7},{"id":"24db2334ffebe478c9f65e231f86ab34744dc48b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-06 10:05:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/7/Nova-ML2-Sriov : SUCCESS in 59m 29s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"9079d88188b96205d3f1d51746dab0f842c2b684","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-06 10:25:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/175780/7/ : SUCCESS in 2h 19m 08s","accounts_in_message":[],"_revision_number":7},{"id":"d91679f9a1988527c59fd1daa7ddee2fc56c0f56","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-12 17:09:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/7 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"bac59468d68b835cccd6d7f3b5782c492f014759","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-07-27 09:18:02.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"160244e10c4f3129a8ba679584f23bb5fbe29db3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-27 09:41:26.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7e44ecec452e475b8d38072298d3270f/index.html : SUCCESS in 12m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5d2801e445e1418bb32a468dff6d3553/index.html : SUCCESS in 15m 07s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b3fbde20cf1244c0938a6772a03a7707/index.html : SUCCESS in 14m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/8/check/real-db-upgrade_nova_percona_user_001:th-percona/8e016a60e87b4ce88e1e92d98f16780c/index.html : SUCCESS in 14m 28s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/004661269c114b6a843657750661a3ba/index.html : SUCCESS in 12m 29s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/8/check/real-db-upgrade_nova_percona_user_002:th-percona/1c32cb5753c24942b37c099ed03b640f/index.html : SUCCESS in 14m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"a2ab9956d0669295495e4227ec5ea11f73121ab5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-27 10:36:50.000000000","message":"Patch Set 8: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/8/5853/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":8},{"id":"78bb8352fc2dda48fe88302dae11748aa5f5da37","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-27 10:40:12.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/8/check-ibm-tempest-dsvm-full/d9bcb80 : SUCCESS in 47m 59s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/8/check-ibm-tempest-dsvm-postgres-full/9bc9f72 : SUCCESS in 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"121c2773c1d864ab4fb5dd69afb2bc3828b70f96","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-27 10:56:10.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/175780/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"3ff980b03c652b51f1b8217151f9d88c04e8ea17","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-27 11:25:56.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/175780/8/ : FAILURE in 1h 18m 45s","accounts_in_message":[],"_revision_number":8},{"id":"3ab89018573a65466d10f18e38fe4158917007b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-27 12:03:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/8/check/gate-nova-pep8/8d2db78/ : SUCCESS in 8m 35s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/8/check/gate-nova-docs/81a803e//doc/build/html/ : SUCCESS in 9m 24s\n- gate-nova-python27 http://logs.openstack.org/80/175780/8/check/gate-nova-python27/2a067ad/ : SUCCESS in 10m 52s\n- gate-nova-python34 http://logs.openstack.org/80/175780/8/check/gate-nova-python34/950ae6c/ : SUCCESS in 13m 20s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-full/220eaf3/ : SUCCESS in 57m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-postgres-full/a1434f1/ : SUCCESS in 38m 41s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-neutron-full/cf861e0/ : SUCCESS in 48m 24s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/8/check/gate-grenade-dsvm/9f66ab9/ : SUCCESS in 47m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-large-ops/b5fd991/ : SUCCESS in 26m 19s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-neutron-large-ops/28c6f58/ : SUCCESS in 35m 45s\n- gate-devstack-dsvm-cells http://logs.openstack.org/80/175780/8/check/gate-devstack-dsvm-cells/c903c11/ : SUCCESS in 22m 04s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/8/check/gate-nova-tox-functional/a6de27a/ : SUCCESS in 13m 19s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/8/check/gate-grenade-dsvm-partial-ncpu/4540b59/ : SUCCESS in 47m 05s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-ironic-pxe_ssh/4d0516b/ : SUCCESS in 37m 37s\n- gate-tempest-dsvm-ironic-pxe_ssh-full-nv http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-ironic-pxe_ssh-full-nv/f67f3a7/ : FAILURE in 2h 18m 39s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-nova-v21-full/aa835e8/ : SUCCESS in 53m 42s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-cells/4f906ae/ : SUCCESS in 35m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-full-ceph/30c84a5/ : SUCCESS in 43m 27s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-multinode-full/4e0f11d/ : FAILURE in 1h 07m 26s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-multinode-smoke/96ad59b/ : SUCCESS in 43m 28s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/80/175780/8/check/gate-tempest-dsvm-neutron-multinode-smoke/be93153/ : SUCCESS in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dcc48732b5950def429394c85e71218c518b099a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-07-27 12:58:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/80/175780/8/check/dsvm-tempest-cinder/bf96271/ : SUCCESS in 42m 21s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"cadbeb68d8d7713e1b8de87ee0eb8edb0112df2a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-29 23:59:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/8 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"0636cf1647777d4538c3654dbbd9d83d2220dc73","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2015-08-04 12:09:21.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"18d48d0a93d5e3cc1bec9e2baa433c1095011c4d","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-08-05 10:52:58.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d096f2374952a74c525bff5dc8294fa11e0231fb","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-08-05 16:26:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"68efd6c877608e3d6f1ca7250c894c9a076d2c2b","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-08-05 16:34:44.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0a846683ffb74e82cf3f030855cf7a7f4ad96322","author":{"_account_id":8276,"name":"kiwik","email":"chenrui.momo@gmail.com","username":"kiwik"},"date":"2015-08-28 04:05:45.000000000","message":"Patch Set 8: Code-Review+1\n\nLGTM. Thank you.","accounts_in_message":[],"_revision_number":8},{"id":"2ca78cf7022115dcb1d3e609049fa3e1e4062fdf","author":{"_account_id":8300,"name":"Abhijeet Malawade","email":"abhijeet.malawade@gmail.com","username":"abhijeet"},"date":"2015-08-28 06:19:08.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"c188f7928839c4d1b486ff0000ca2a57f05fbd3a","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-08-28 06:45:45.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"649595d6aa7eccdd70d31897e9d548626a0b0ffa","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-08-28 09:06:37.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4d60ef8d5e648b6c2ac7beddac56f232a629193f","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-08-31 07:50:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3ba24dd4325e27c1f6a8c4a9206c1b3c6e5f867c","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-08-31 08:54:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f6accad7a06fb835575bd52d7ebdad0069d8d4cd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-31 09:30:52.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/00bc0580335e4488b0b8f03bdab37fdf/index.html : SUCCESS in 14m 21s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/659ccf6896e7476bbd9d936537aac960/index.html : SUCCESS in 15m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dad641c0fb294240b526e13c2b53cc01/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/9/check/real-db-upgrade_nova_percona_user_001:th-percona/8760812972854ecfa497ba7bec1781cd/index.html : SUCCESS in 16m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/304eeacdcbd2459e97afb7b3d5753051/index.html : SUCCESS in 14m 15s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/9/check/real-db-upgrade_nova_percona_user_002:th-percona/788aba1333d04966be5fd3578906ad93/index.html : SUCCESS in 11m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"f06b4f094833f1bfcf9d72a9c7cf4413a7ab2cdd","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-08-31 09:34:15.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"754b77732dfe4a885ac3fe0069c6b4739705087e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-31 09:53:31.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/175780/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"39aa723d57f577316177e405f00301c4e9a1696d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-31 09:56:34.000000000","message":"Patch Set 9: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/9/10388/index.html : SUCCESS in 53 m 45 s","accounts_in_message":[],"_revision_number":9},{"id":"0bc56f579b1e1992f3ac49df2bec0ecb257213fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-31 10:09:29.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/9/check-ibm-tempest-dsvm-full/f79a9aa : SUCCESS in 1h 14m 10s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/9/check-ibm-tempest-dsvm-postgres-full/3d6d58b : SUCCESS in 48m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fa5b89c780957b149bc4068a81d3f9f97d3bb097","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-31 10:11:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/9/check/gate-nova-pep8/211699f/ : SUCCESS in 7m 21s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/9/check/gate-nova-docs/dd01154//doc/build/html/ : SUCCESS in 10m 57s\n- gate-nova-python27 http://logs.openstack.org/80/175780/9/check/gate-nova-python27/3758a03/ : SUCCESS in 13m 51s\n- gate-nova-python34 http://logs.openstack.org/80/175780/9/check/gate-nova-python34/9c6bd52/ : SUCCESS in 10m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-full/cf279e0/ : SUCCESS in 57m 42s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-postgres-full/1ea0697/ : SUCCESS in 46m 16s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-neutron-full/4d70d8c/ : SUCCESS in 53m 26s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/9/check/gate-grenade-dsvm/6267d94/ : SUCCESS in 39m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-large-ops/3ae426c/ : SUCCESS in 26m 04s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-neutron-large-ops/e10ad5d/ : SUCCESS in 26m 10s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/9/check/gate-nova-tox-functional/0ac6ad9/ : SUCCESS in 13m 27s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/9/check/gate-grenade-dsvm-partial-ncpu/7c2df91/ : SUCCESS in 39m 29s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-ironic-pxe_ssh/853d9f8/ : SUCCESS in 33m 33s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-nova-v21-full/addff31/ : SUCCESS in 56m 46s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-cells/e10fd0c/ : SUCCESS in 37m 32s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-full-ceph/ab6ec43/ : SUCCESS in 41m 12s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/9/check/gate-tempest-dsvm-multinode-full/afe7836/ : SUCCESS in 49m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9ee7a9962cbbb859a78e5dbbb2f4f1ee0addb89e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-31 10:12:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/9/check/dsvm-tempest-xen/b9509b8 : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":9},{"id":"2a6437f1a1b2e898d254a6064ff3af2a4cd8cb89","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-08-31 10:39:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"af3fa13dfec7849733dff3927019d8aec37a0f2f","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-08-31 10:41:03.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"364bba5873723a9f50a4efc0419957f9888ae9b2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-31 10:42:50.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/175780/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"bf1fdb94b189bbc4109c777ddb0fb3f3b40838be","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-08-31 10:47:11.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"10e1e827bc75734698ff92d37eb778ddd5896caf","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-31 10:57:03.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/175780/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"8e4db60277c18393933f329c1fd70e66010bf832","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-31 11:07:34.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/fecd0342a0ce4d06862ba263b8a7c732/index.html : SUCCESS in 14m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f8ebb6b975d84c8d9b7a1aedf9df2152/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/bde6a56b4bb3478cb17b557874166327/index.html : SUCCESS in 11m 45s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/10/check/real-db-upgrade_nova_percona_user_001:th-percona/cffd83f2439844fda4803fe021c55351/index.html : SUCCESS in 15m 56s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f37733a9343c479885c07efd650517ef/index.html : SUCCESS in 12m 57s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/10/check/real-db-upgrade_nova_percona_user_002:th-percona/2d12b4d3f46b4256add52e4471530828/index.html : SUCCESS in 12m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"e77f707eb41cabe59ad9da85284a68e5050b1ae9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-31 11:26:52.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/10/check-ibm-tempest-dsvm-full/e287001 : SUCCESS in 47m 17s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/10/check-ibm-tempest-dsvm-postgres-full/c09526b : SUCCESS in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fbcec483651b62bb6da54a9af9bf75e7daa9f40e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-31 11:39:56.000000000","message":"Patch Set 10: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/10/10400/index.html : SUCCESS in 57 m 44 s","accounts_in_message":[],"_revision_number":10},{"id":"160c7cda4ceb08c1cf4f6cc1d86276e2cb56646a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-31 12:03:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/10/check/dsvm-tempest-xen/2174cf7 : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":10},{"id":"d12eaa01641b5b32d153d1599e789d903a7fe1b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-31 12:34:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/10/check/gate-nova-pep8/3c79544/ : SUCCESS in 8m 46s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/10/check/gate-nova-docs/aaef361//doc/build/html/ : SUCCESS in 14m 08s\n- gate-nova-python27 http://logs.openstack.org/80/175780/10/check/gate-nova-python27/0234178/ : SUCCESS in 14m 13s\n- gate-nova-python34 http://logs.openstack.org/80/175780/10/check/gate-nova-python34/07f65bf/ : SUCCESS in 9m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-full/695aa72/ : SUCCESS in 42m 38s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-postgres-full/64a9d1b/ : SUCCESS in 54m 08s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-neutron-full/8cef334/ : SUCCESS in 1h 12m 25s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/10/check/gate-grenade-dsvm/5ad80f9/ : SUCCESS in 49m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-large-ops/863fb57/ : SUCCESS in 38m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-neutron-large-ops/e148838/ : SUCCESS in 34m 43s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/10/check/gate-nova-tox-functional/08a2d2b/ : SUCCESS in 14m 29s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/10/check/gate-grenade-dsvm-partial-ncpu/15b0a0f/ : SUCCESS in 49m 39s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-ironic-pxe_ssh/41c14f3/ : SUCCESS in 50m 30s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-nova-v21-full/7a2a734/ : SUCCESS in 1h 03m 56s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-cells/ea2a9d1/ : FAILURE in 52m 36s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-full-ceph/a9866f5/ : SUCCESS in 45m 45s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-multinode-full/97d3cc4/ : FAILURE in 57m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"759ffadb51534c8ac7a8e0c4b77cb7aa81718ec7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-08-31 12:35:15.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells: https://bugs.launchpad.net/bugs/1489581\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"0d82e8394df455a10bff7fc3dc3a344bf90a3ce7","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-09-01 09:03:04.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"b156c8f3e418d2bfc9a6f35aefdcb2443ed5cfd3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-01 09:55:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/10 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"253502a67d90246cfb9f3d843be91702f0a3cdfd","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-09-01 10:34:43.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"4bb2696f9fe4b47bf43913e9317c47c85a62adb7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-01 11:53:53.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/10/check/gate-nova-pep8/5df7323/ : SUCCESS in 10m 21s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/10/check/gate-nova-docs/2fba7c4//doc/build/html/ : SUCCESS in 17m 56s\n- gate-nova-python27 http://logs.openstack.org/80/175780/10/check/gate-nova-python27/0dd766c/ : SUCCESS in 16m 21s\n- gate-nova-python34 http://logs.openstack.org/80/175780/10/check/gate-nova-python34/1ebeb2b/ : SUCCESS in 19m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-full/1c03afa/ : SUCCESS in 58m 53s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-postgres-full/f3479ab/ : SUCCESS in 56m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-neutron-full/d972fbd/ : SUCCESS in 1h 16m 03s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/10/check/gate-grenade-dsvm/1ce9d19/ : SUCCESS in 51m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-large-ops/102f387/ : SUCCESS in 43m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-neutron-large-ops/c0c6ae7/ : SUCCESS in 42m 21s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/10/check/gate-nova-tox-functional/26062ae/ : SUCCESS in 19m 50s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/10/check/gate-grenade-dsvm-partial-ncpu/2989033/ : SUCCESS in 58m 11s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-ironic-pxe_ssh/a294b83/ : FAILURE in 34m 57s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-nova-v21-full/a3256e7/ : SUCCESS in 57m 12s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-cells/ba31946/ : SUCCESS in 55m 06s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-full-ceph/5da2328/ : SUCCESS in 1h 00m 25s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/10/check/gate-tempest-dsvm-multinode-full/24912ee/ : FAILURE in 1h 08m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7c3b0b57ff6fd3cf897ca59acb81f941b693c91a","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-09-04 08:01:35.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8054fb3a181983c8c7e6af645dc0498d7986f5ff","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-04 08:02:22.000000000","message":"Patch Set 10: Verified+1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"ef497d4dc6059132c1ad7cd16a566221f5d1b421","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-04 08:10:16.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"6835e9807b418300bfaffcef28b86a88b273c1ee","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-09-04 08:46:07.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"94e9e219c93911c3801d00932e42e90f8addd6a0","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-09-04 08:53:39.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"3967ae7c2fea666188e341ee5daa461cf7a45443","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-04 09:07:09.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/175780/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"9310e1abff8e9c5196691f5175c300bb4e666cae","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-04 09:08:33.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/246ca711c9294a4b83ebdde637fe38ba/index.html : SUCCESS in 15m 35s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/92fdab7e290f4f9b9d946aedda605b57/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/32e4d9d9b0d54f7e8bc28f32eeaff9a6/index.html : SUCCESS in 14m 34s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/11/check/real-db-upgrade_nova_percona_user_001:th-percona/06948a58865c4a75b200d7007a65783b/index.html : SUCCESS in 17m 22s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ea5eca9fd8764ad9a506b2a47058954c/index.html : SUCCESS in 13m 47s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/11/check/real-db-upgrade_nova_percona_user_002:th-percona/482b6ad3e7da4d0b93780a9717a1bbf3/index.html : SUCCESS in 14m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"eccf27fd3fe76079b1ce97d7d816c50e30a64675","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-04 09:41:38.000000000","message":"Patch Set 11: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/11/10928/index.html : SUCCESS in 53 m 32 s","accounts_in_message":[],"_revision_number":11},{"id":"37525ec4e8013d0e553067d9cfa89318f23aad64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-04 09:55:59.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/11/check/gate-nova-pep8/676372b/ : SUCCESS in 8m 34s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/11/check/gate-nova-docs/d5379ee//doc/build/html/ : SUCCESS in 15m 04s\n- gate-nova-python27 http://logs.openstack.org/80/175780/11/check/gate-nova-python27/e54d34d/ : FAILURE in 14m 59s\n- gate-nova-python34 http://logs.openstack.org/80/175780/11/check/gate-nova-python34/bf879bf/ : SUCCESS in 14m 15s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-full/c0c4222/ : SUCCESS in 43m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-postgres-full/ef3a6c1/ : SUCCESS in 46m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-neutron-full/97d5d5f/ : SUCCESS in 1h 06m 35s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/11/check/gate-grenade-dsvm/ef8296a/ : SUCCESS in 43m 54s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-large-ops/aa6a149/ : FAILURE in 33m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-neutron-large-ops/5bf2aa2/ : SUCCESS in 33m 29s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/11/check/gate-nova-tox-functional/467bb71/ : SUCCESS in 14m 18s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/11/check/gate-grenade-dsvm-partial-ncpu/ee6aa66/ : SUCCESS in 48m 51s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/37084c4/ : SUCCESS in 49m 15s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-nova-v21-full/5c5baba/ : SUCCESS in 52m 55s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-cells/4329f7d/ : SUCCESS in 46m 42s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-full-ceph/c2b7a43/ : SUCCESS in 41m 24s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/11/check/gate-tempest-dsvm-multinode-full/4139fd5/ : SUCCESS in 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a9acf208f18e599281ab379563eece680c8abd75","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-04 10:04:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/11/check/dsvm-tempest-xen/957b35c : FAILURE in 1h 17m 40s","accounts_in_message":[],"_revision_number":11},{"id":"6919767fb442ea3424c322fd13fb648bedad0829","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-04 10:12:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/175780/11/ : FAILURE in 1h 25m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e84b9f1da75d71a7c2aeeaedcdd8675b3116e176","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-04 13:34:39.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/11/check-ibm-tempest-dsvm-full/c38608a : SUCCESS in 51m 03s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/11/check-ibm-tempest-dsvm-postgres-full/98a1ecb : SUCCESS in 47m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7ccb5ad2bf9fd8011a4eddd63539856e759bab7f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-05 00:56:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/11 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"32344ca22a905d2a447f86f8bd2478cc99cea0cf","author":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"date":"2015-09-06 12:00:29.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"fae9d0d84ed2e26f5034893f583dbc4da0b3b7a3","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-09-18 08:30:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9f5ea3daca4ebb4c76be922353e759581790ca3f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-18 09:00:06.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3fa817c80d4948c4afd4426065ec8752/index.html : SUCCESS in 12m 12s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1887bbda57b945ffb85ec9d8f9524c45/index.html : SUCCESS in 16m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/466ff61841b54cbb835c58655b98ef25/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/12/check/real-db-upgrade_nova_percona_user_001:th-percona/3ef528ec5f154d5ca62fa5e1041e3845/index.html : SUCCESS in 13m 06s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4ae48185ba494b829ef1548e563638f1/index.html : SUCCESS in 15m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/12/check/real-db-upgrade_nova_percona_user_002:th-percona/97aead4b65fd4bb89dddeff54ee2d361/index.html : SUCCESS in 12m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"82fe323c2be8d339f23e3e6d35939ef530cb0eaf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-18 09:19:34.000000000","message":"Patch Set 12:\n\nBuild failed. Logs: http://208.91.1.172/logs/175780/12 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"33455ad7e73f26d6e05fd368d5c178588ba12f7f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-18 09:36:45.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/175780/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"9886674c8b1536528b749d15399f6b12dc510bc4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-18 09:53:27.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/12/check/gate-nova-pep8/13a93ea/ : SUCCESS in 8m 53s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/12/check/gate-nova-docs/4e0972b//doc/build/html/ : SUCCESS in 14m 19s\n- gate-nova-python27 http://logs.openstack.org/80/175780/12/check/gate-nova-python27/fb49152/ : FAILURE in 15m 03s\n- gate-nova-python34 http://logs.openstack.org/80/175780/12/check/gate-nova-python34/ac6fc2d/ : SUCCESS in 13m 08s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-full/363bd49/ : SUCCESS in 53m 20s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-postgres-full/959775b/ : SUCCESS in 55m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-neutron-full/aa66d81/ : SUCCESS in 1h 19m 48s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/12/check/gate-grenade-dsvm/dcdfa9c/ : SUCCESS in 49m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-large-ops/e40fba8/ : SUCCESS in 33m 45s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-neutron-large-ops/65c70c7/ : SUCCESS in 35m 02s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/12/check/gate-nova-tox-functional/687e549/ : SUCCESS in 16m 24s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/12/check/gate-grenade-dsvm-partial-ncpu/54aa86a/ : SUCCESS in 55m 43s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/64f9595/ : FAILURE in 43m 41s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-nova-v20-api/ca36a20/ : SUCCESS in 45m 40s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-nova-v20-api-legacy/cf68519/ : SUCCESS in 41m 25s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-cells/348a52e/ : SUCCESS in 59m 23s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-full-ceph/213d019/ : SUCCESS in 54m 47s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/12/check/gate-tempest-dsvm-multinode-full/045c707/ : SUCCESS in 48m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6083579bfa5d5d79e3da3029511fb87aa13d5a69","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-18 10:13:37.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/12/Nova-ML2-Sriov : SUCCESS in 50m 18s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"41689419ef347c14d9bcc561a188c8bf9d2cb96c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-18 11:25:32.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/175780/12/ : FAILURE in 2h 54m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a62b8ae4b24fb7d8d55f85973ad4bcccde509431","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-18 11:27:25.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/12/check-ibm-tempest-dsvm-full/c05e1db : SUCCESS in 56m 13s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/12/check-ibm-tempest-dsvm-postgres-full/93b3ead : SUCCESS in 54m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"72996f81065c455f45c299cb7dc39da1ba9d1af3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-09-18 12:37:09.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"b0f78a66bdac2cb7c340a5d711cf6d1d5485f3f8","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-09-18 12:38:20.000000000","message":"Patch Set 12: -Code-Review\n\noops, the unittest failed.","accounts_in_message":[],"_revision_number":12},{"id":"1dfe9e121b82379a6f2c020074d2f296cff7e75c","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-09-21 15:48:40.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3175e062a9f6807c2c7e741f9ef8c12b52857f16","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2015-09-22 02:26:07.000000000","message":"Patch Set 12: Code-Review-1\n\nCodes looks good to me, but unit test error should be fixed.","accounts_in_message":[],"_revision_number":12},{"id":"c754df8ac240162385a703f4a39094b5c5b0c193","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-09-22 10:19:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1ef3fe7af87165c6becdf3fb4772b3b2e06eee08","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-22 10:51:51.000000000","message":"Patch Set 13: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/13/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/05b0927e4fc94e52baa36f79e9c48975/index.html : SUCCESS in 14m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/13/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ce568a2a34f64222b273d818b783d91d/index.html : SUCCESS in 16m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/13/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/751e106cd13c440abab09dcb461f0e03/index.html : SUCCESS in 11m 58s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/13/check/real-db-upgrade_nova_percona_user_001:th-percona/0010e0e38b4042edb45bdc740f639a5c/index.html : SUCCESS in 15m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/13/check/real-db-upgrade_nova_mysql_user_002:th-mysql/909d88940a83405c8e1c3df252471f18/index.html : SUCCESS in 14m 20s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/13/check/real-db-upgrade_nova_percona_user_002:th-percona/5d92bae9cae247dcb27c2a6a99ec0cec/index.html : SUCCESS in 12m 14s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":13},{"id":"c4a0bd5e92e7e23e216820b4d67a88f1d443bd80","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-22 11:13:41.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/13/check-ibm-tempest-dsvm-full/3b9e8cf : SUCCESS in 53m 35s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175780/13/check-ibm-tempest-dsvm-postgres-full/44a911b : SUCCESS in 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b2f57a74f6fcb9514cfe5f306c372e55cab1b76f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-22 11:17:31.000000000","message":"Patch Set 13: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/13/12019/index.html : SUCCESS in 52 m 22 s","accounts_in_message":[],"_revision_number":13},{"id":"8ed5e2b4606325c4d235180b420547c661cb9deb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-22 11:18:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/13/check/gate-nova-pep8/1fb1a24/ : SUCCESS in 9m 17s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/13/check/gate-nova-docs/72d7026//doc/build/html/ : SUCCESS in 14m 52s\n- gate-nova-python27 http://logs.openstack.org/80/175780/13/check/gate-nova-python27/3918a8f/ : SUCCESS in 13m 40s\n- gate-nova-python34 http://logs.openstack.org/80/175780/13/check/gate-nova-python34/575f5a0/ : SUCCESS in 10m 17s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-full/3f09c08/ : SUCCESS in 42m 30s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-postgres-full/d1101b3/ : SUCCESS in 43m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-neutron-full/42d3e6a/ : SUCCESS in 57m 42s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/13/check/gate-grenade-dsvm/f571cb7/ : SUCCESS in 49m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-large-ops/18baa62/ : SUCCESS in 25m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-neutron-large-ops/c25062e/ : SUCCESS in 31m 42s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/13/check/gate-nova-tox-functional/79b698b/ : SUCCESS in 16m 43s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/13/check/gate-grenade-dsvm-partial-ncpu/a8cf8eb/ : SUCCESS in 46m 33s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/767cd10/ : SUCCESS in 45m 42s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-nova-v20-api/b9bc050/ : SUCCESS in 33m 13s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-nova-v20-api-legacy/6809688/ : SUCCESS in 44m 49s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-cells/66156ba/ : SUCCESS in 52m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-full-ceph/a8fc519/ : SUCCESS in 50m 36s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/13/check/gate-tempest-dsvm-multinode-full/8b00c3d/ : SUCCESS in 48m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9c17101da9d1fa8337df6e97d797daa332db5544","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-22 11:56:49.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/80/175780/13/check/check-dsvm-vz7-tempest-exe-minimal/9cd85dc/ : SUCCESS in 35m 18s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"6f9ac60ce1f8bd4c1ee9a21f120f36f8d308ebac","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-22 12:03:27.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/80/175780/13/check/dsvm-tempest-cinder/3e417c8/ : SUCCESS in 42m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"f464602cfb1a111d1aa3dacb0cacc5846e514669","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-22 12:42:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/13/Nova-ML2-Sriov : SUCCESS in 43m 11s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"8d7cea1584fbfbf1771dee26b708e30bfb782702","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-22 13:44:58.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/175780/13/ : SUCCESS in 1h 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bce77c92d66b6a5f5c02b75a03fc50d0e59a4f6d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-22 18:53:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/13 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"35a64acfafded8596f206210fc67ef1913accbbc","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-09-29 19:44:33.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"ff7cd9587c82202b84af98022544a05fcc4849d2","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-10-06 15:08:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"8fbd577a8ff883f09d9569c541f8ec9f365f883f","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-10-06 16:02:17.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"98af55350fd5af99a4a6ec14170b5af49d36e04c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-06 16:12:43.000000000","message":"Patch Set 14: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/14/13019/index.html : SUCCESS in 59 m 11 s","accounts_in_message":[],"_revision_number":14},{"id":"6613092dc192e6c3b6779f4eeac016eb5c138a60","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-06 16:25:37.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/14/check/dsvm-tempest-xen/edb28d3 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":14},{"id":"2cab6de79da7596e6b9cb1bb7ebeb83afab5b1d8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-06 16:50:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/14/check/gate-nova-pep8/5f2917d/ : SUCCESS in 7m 02s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/14/check/gate-nova-docs/4b0e611//doc/build/html/ : SUCCESS in 15m 01s\n- gate-nova-python27 http://logs.openstack.org/80/175780/14/check/gate-nova-python27/3356a73/ : SUCCESS in 15m 30s\n- gate-nova-python34 http://logs.openstack.org/80/175780/14/check/gate-nova-python34/e898b0c/ : SUCCESS in 13m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-full/6d08efb/ : FAILURE in 50m 21s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-postgres-full/516082a/ : FAILURE in 1h 00m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-neutron-full/1e3b5e4/ : SUCCESS in 1h 26m 01s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/14/check/gate-grenade-dsvm/56df976/ : SUCCESS in 55m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-large-ops/a7eb560/ : SUCCESS in 42m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-neutron-large-ops/9fa1fa2/ : SUCCESS in 40m 27s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/14/check/gate-nova-tox-functional/76538c3/ : SUCCESS in 13m 00s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/14/check/gate-grenade-dsvm-partial-ncpu/97fb202/ : SUCCESS in 56m 21s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2f895b6/ : SUCCESS in 46m 45s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-nova-v20-api/e1409e6/ : SUCCESS in 35m 08s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-nova-v20-api-legacy/1e563f4/ : SUCCESS in 48m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-cells/52e3a30/ : SUCCESS in 52m 02s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-full-ceph/ad24d89/ : SUCCESS in 59m 39s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/14/check/gate-tempest-dsvm-multinode-full/fdac45f/ : SUCCESS in 1h 09m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"15fd028a283a853a706c9f9c817a0540d7e4a0d5","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-06 16:51:59.000000000","message":"Patch Set 14:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1501558\n- gate-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1501366\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":14},{"id":"ef132e19da6d76217574b8d7f10a2de897909e66","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-06 17:29:44.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/175780/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"2ee455d626efcb54654a4c5d08e00b4b0aa98ae9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-06 18:17:14.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/14/check/check-ibm-tempest-dsvm-full/5a895aa/ : SUCCESS in 48m 04s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/14/check/check-ibm-tempest-dsvm-postgres-full/b34f4fb/ : SUCCESS in 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ea29beb216a9fa6c738f7fa2bc5c87dc4b9c641f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-06 19:31:28.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/14/ : FAILURE in 2h 38m 30s","accounts_in_message":[],"_revision_number":14},{"id":"44edcfa205b427671a119e7baa00ab273c397ecd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-06 23:06:48.000000000","message":"Patch Set 14: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/14 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"faa9b689efce5db35783cb4f887c2b9659d79014","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-07 03:39:00.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/14/Nova-ML2-Sriov : SUCCESS in 47m 52s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":14},{"id":"06c2d7446f2b4fd17f30cb3463f48b2a4b921538","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-07 16:27:02.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/24ad7eca65954ce19dd0a5b76d01e059/index.html : SUCCESS in 15m 22s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d7b017518f994e0d9ba39d071a0c7e8d/index.html : SUCCESS in 18m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fc7f57ed14e749ceb12469adcd430b2e/index.html : SUCCESS in 15m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/14/check/real-db-upgrade_nova_percona_user_001:th-percona/74bd969419a84015882b9000e21f9342/index.html : SUCCESS in 17m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/14/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4f5ba40fa30c4c57babe360b50a8d751/index.html : SUCCESS in 14m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/14/check/real-db-upgrade_nova_percona_user_002:th-percona/0145f556def14fe8bed40d7fc5119884/index.html : SUCCESS in 14m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"385bd421bf18e93a3322e67f2e2c1952b7bec22f","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-11-04 08:28:17.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"5162662abdf11b74facfd46a05d6fa11b26f002d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-04 08:46:03.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/175780/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"e81d684ad147054c7a3879ed89ee47466c058ba4","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-04 08:53:44.000000000","message":"Patch Set 15: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/15/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0c089054387b41eb97204431bd4cf29c/index.html : SUCCESS in 14m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/15/check/real-db-upgrade_nova_mysql_user_001:th-mysql/50cfe6832dba48d4b5e43ae18af17920/index.html : SUCCESS in 15m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/15/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/44aa1fa400f2434bad2d3ce67c201f54/index.html : SUCCESS in 14m 06s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/15/check/real-db-upgrade_nova_percona_user_001:th-percona/68525b13d09741d59bce9440bf0d02a3/index.html : SUCCESS in 16m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/15/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c0e171bea424ab9a615bb8e1feb275c/index.html : SUCCESS in 14m 51s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/15/check/real-db-upgrade_nova_percona_user_002:th-percona/b97a1f3e505a4e9989c446d09b6fd5f9/index.html : SUCCESS in 12m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":15},{"id":"a0a221e69aa69432ccdea3b9f58d632945c323b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-04 09:35:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/15/check/gate-nova-pep8/b9317e1/ : SUCCESS in 8m 07s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/15/check/gate-nova-docs/42aa1d4//doc/build/html/ : SUCCESS in 10m 26s\n- gate-nova-python27 http://logs.openstack.org/80/175780/15/check/gate-nova-python27/255b549/ : SUCCESS in 13m 35s\n- gate-nova-python34 http://logs.openstack.org/80/175780/15/check/gate-nova-python34/d333317/ : SUCCESS in 13m 45s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full/a0ee4db/ : SUCCESS in 42m 15s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-postgres-full/4fbeb67/ : SUCCESS in 55m 20s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-full/df7670e/ : SUCCESS in 58m 41s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm/58f0776/ : SUCCESS in 51m 01s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-large-ops/d71b010/ : SUCCESS in 37m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-large-ops/4c85f79/ : SUCCESS in 27m 36s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/15/check/gate-nova-tox-functional/f3904c1/ : SUCCESS in 16m 06s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-partial-ncpu/1dd05d3/ : SUCCESS in 49m 06s\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-multinode-nv/c836817/ : SUCCESS in 53m 28s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c5ed21e/ : SUCCESS in 47m 54s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api/eff1de1/ : FAILURE in 44m 41s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api-legacy/bc7ca06/ : SUCCESS in 45m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-cells/b9ecfef/ : SUCCESS in 55m 58s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full-ceph/794a059/ : SUCCESS in 45m 19s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-multinode-full/be09c0b/ : FAILURE in 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ef5441b5f80c4a72e71a2a3ce145ea8740a271e8","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-11-04 09:36:04.000000000","message":"Patch Set 15:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-nova-v20-api: https://bugs.launchpad.net/bugs/1501558\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":15},{"id":"e4316b3eb7a5031c9e0fce3104e28174008e821b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-04 09:44:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/15/check/dsvm-tempest-xen/77d3c16 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":15},{"id":"926ee64165287ce3ef100932a0b9af433b95239b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-04 09:46:36.000000000","message":"Patch Set 15: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/15/14857/index.html : SUCCESS in 01 h 14 m 23 s","accounts_in_message":[],"_revision_number":15},{"id":"117ebc5562b6f52e89e4124f90170be4b5c021e8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-04 11:28:50.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/15/ : FAILURE in 3h 00m 01s","accounts_in_message":[],"_revision_number":15},{"id":"ec55a8ed55a6b904be83d5a9b2f857e20c410735","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-11-04 11:41:08.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/15/check/check-ibm-tempest-dsvm-full/8a6293f/ : SUCCESS in 56m 21s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/15/check/check-ibm-tempest-dsvm-postgres-full/651f927/ : SUCCESS in 51m 57s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"86a1e632c4f017bbd57b604174b9b080decb4e31","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-11-04 13:55:34.000000000","message":"Patch Set 15:\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/80/175780/15/compute-nfv-flavors : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"180abc38149a79d678974d80de9078e5d69586be","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2015-11-05 07:23:28.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"c772bd0cefa5f516fd6506cae673751f8e021a3a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-05 08:21:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/15/ : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":15},{"id":"0ebfcd3a1993c00b3fd79007b59bc9fad91f17d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-05 08:33:29.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/15/check/gate-nova-pep8/f62d122/ : SUCCESS in 6m 42s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/15/check/gate-nova-docs/31fe041//doc/build/html/ : SUCCESS in 8m 50s\n- gate-nova-python27 http://logs.openstack.org/80/175780/15/check/gate-nova-python27/e508f4e/ : SUCCESS in 13m 31s\n- gate-nova-python34 http://logs.openstack.org/80/175780/15/check/gate-nova-python34/21b759c/ : SUCCESS in 12m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full/99c370a/ : SUCCESS in 53m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-postgres-full/86f13cd/ : SUCCESS in 50m 38s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-full/e8817f0/ : SUCCESS in 1h 09m 46s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm/cc50e99/ : SUCCESS in 58m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-large-ops/cdc0693/ : SUCCESS in 37m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-large-ops/fd4ba89/ : SUCCESS in 29m 35s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/15/check/gate-nova-tox-functional/15c38da/ : SUCCESS in 15m 49s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-partial-ncpu/88a6060/ : SUCCESS in 44m 17s\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-multinode-nv/cff41dd/ : SUCCESS in 51m 42s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/fffb7fb/ : SUCCESS in 1h 00m 11s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api/a882aae/ : SUCCESS in 41m 30s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api-legacy/8d3e4ba/ : SUCCESS in 43m 17s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-cells/d8ece2a/ : FAILURE in 39m 26s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full-ceph/df21a43/ : SUCCESS in 53m 05s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-multinode-full/eef4435/ : SUCCESS in 58m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"be9f8e4836be91e10297fe38b5392d87172d9267","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-11-05 08:33:41.000000000","message":"Patch Set 15:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells: https://bugs.launchpad.net/bugs/1489581\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":15},{"id":"deed19dd3259de404230744ff394ff68bf77416e","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2015-11-05 08:47:48.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"4a4551ec5eb0f464dc89f53d790c81fde0ede194","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-05 09:58:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/15/ : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":15},{"id":"647e5fcee09ed87480887eb90f46ab52ec790591","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-05 10:08:07.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/15/check/gate-nova-pep8/bb41c23/ : SUCCESS in 6m 30s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/15/check/gate-nova-docs/ccbf169//doc/build/html/ : SUCCESS in 14m 36s\n- gate-nova-python27 http://logs.openstack.org/80/175780/15/check/gate-nova-python27/061c891/ : SUCCESS in 14m 11s\n- gate-nova-python34 http://logs.openstack.org/80/175780/15/check/gate-nova-python34/0c119cf/ : SUCCESS in 13m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full/92d8597/ : SUCCESS in 55m 53s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-postgres-full/e693b2b/ : SUCCESS in 40m 19s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-full/eff5691/ : SUCCESS in 1h 20m 06s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm/8a5f800/ : SUCCESS in 46m 45s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-large-ops/903fc13/ : SUCCESS in 39m 16s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-large-ops/97af5d4/ : SUCCESS in 37m 16s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/15/check/gate-nova-tox-functional/e9bf178/ : SUCCESS in 16m 53s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-partial-ncpu/20ad02b/ : SUCCESS in 45m 43s\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-multinode-nv/4ecb79d/ : FAILURE in 8m 43s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/7f3e4a7/ : SUCCESS in 39m 09s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api/932e0ad/ : SUCCESS in 31m 09s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api-legacy/d491e3d/ : SUCCESS in 31m 28s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-cells/7ed806e/ : FAILURE in 38m 33s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full-ceph/8fbf0be/ : FAILURE in 1h 03m 29s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-multinode-full/bb6a0d3/ : SUCCESS in 1h 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"86ad1ebe95153dd794d9d773d21ac8663f392cf2","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-11-05 10:08:22.000000000","message":"Patch Set 15:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells: https://bugs.launchpad.net/bugs/1489581\n- gate-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1501558\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":15},{"id":"d1a41156bf63da2996f0262737bfa7c96d75e6e5","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2015-11-05 10:09:16.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"26b3ef0efc7fecd565b94fcb2033a3f8046e5a4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-05 11:40:54.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/15/check/gate-nova-pep8/7f4aebb/ : SUCCESS in 6m 45s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/15/check/gate-nova-docs/bd6a293//doc/build/html/ : SUCCESS in 14m 54s\n- gate-nova-python27 http://logs.openstack.org/80/175780/15/check/gate-nova-python27/0c4a125/ : SUCCESS in 13m 58s\n- gate-nova-python34 http://logs.openstack.org/80/175780/15/check/gate-nova-python34/926971e/ : SUCCESS in 13m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full/83c7bac/ : SUCCESS in 43m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-postgres-full/7dbeb5c/ : SUCCESS in 47m 00s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-full/df4d957/ : SUCCESS in 1h 10m 12s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm/96a3ab4/ : SUCCESS in 52m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-large-ops/ad5e749/ : SUCCESS in 38m 19s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-neutron-large-ops/e5fd25d/ : SUCCESS in 33m 30s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/15/check/gate-nova-tox-functional/7a6815a/ : SUCCESS in 16m 04s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-partial-ncpu/0b24935/ : SUCCESS in 42m 56s\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/80/175780/15/check/gate-grenade-dsvm-multinode-nv/6457b4c/ : SUCCESS in 53m 38s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a529392/ : SUCCESS in 40m 52s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api/53c50c6/ : SUCCESS in 39m 43s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-nova-v20-api-legacy/645d589/ : SUCCESS in 43m 22s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-cells/4c489eb/ : FAILURE in 38m 19s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-full-ceph/9864a37/ : SUCCESS in 42m 47s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/15/check/gate-tempest-dsvm-multinode-full/da36e26/ : SUCCESS in 1h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"fd7ef0cbc130c48343ea159f290fed78e553e30f","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-11-05 11:41:06.000000000","message":"Patch Set 15:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells: https://bugs.launchpad.net/bugs/1489581\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":15},{"id":"b870183e9c9ca3a333d70addae538371625475a3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-05 12:32:35.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/15/ : FAILURE in 1h 50m 13s","accounts_in_message":[],"_revision_number":15},{"id":"1bd83e786344cacd9bc788177a81f1bc03e3e346","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-06 11:35:01.000000000","message":"Patch Set 15: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/15 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"1defcd1e88aae0d0e0360443edccc5bcf074b59b","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-11-16 09:59:35.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c404d91f3e5e240ae7ae74232cb334f73944ae5a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-16 10:15:23.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/175780/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"d5787c9ec88d085811d002cf87c82cccfcb53333","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-16 10:49:08.000000000","message":"Patch Set 16: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/16/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8d61702b2b1b4f5298366d99b3b0c9b0/index.html : SUCCESS in 16m 28s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/16/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8c7c8e57070d49d59d0d06096dc5ed09/index.html : SUCCESS in 16m 40s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/16/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/403b3a07fb704d5c939c53461a9a21b7/index.html : SUCCESS in 14m 36s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/16/check/real-db-upgrade_nova_percona_user_001:th-percona/f347fcdd8f244393b3ea55612a49d3e1/index.html : SUCCESS in 16m 40s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/16/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c9922e416fb346219512bab735c0aade/index.html : SUCCESS in 14m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/16/check/real-db-upgrade_nova_percona_user_002:th-percona/5d6dc5a9818c4afcbf0fc89267ed285d/index.html : SUCCESS in 15m 34s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":16},{"id":"e64c4ac72a0cbd62c90ec4336875584e75408a81","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-11-16 10:53:34.000000000","message":"Patch Set 16:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/16/check/check-ibm-tempest-dsvm-full/074be25/ : SUCCESS in 53m 27s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/16/check/check-ibm-tempest-dsvm-postgres-full/a7c54e9/ : SUCCESS in 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"2f327cf31c228dff8b05733f3a78ef7f9e2934d5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-16 10:58:39.000000000","message":"Patch Set 16: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/16/16068/index.html : SUCCESS in 56 m 01 s","accounts_in_message":[],"_revision_number":16},{"id":"ac6bd99104d43817b74320cd5502f8282e7a44b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-16 11:09:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/16/check/gate-nova-pep8/9bf5037/ : SUCCESS in 9m 27s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/16/check/gate-nova-docs/fc27325//doc/build/html/ : SUCCESS in 14m 06s\n- gate-nova-python27 http://logs.openstack.org/80/175780/16/check/gate-nova-python27/e0fc262/ : SUCCESS in 14m 27s\n- gate-nova-python34 http://logs.openstack.org/80/175780/16/check/gate-nova-python34/a42f670/ : SUCCESS in 13m 27s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-full/35cb5ca/ : SUCCESS in 56m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-postgres-full/99876a2/ : SUCCESS in 50m 41s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-neutron-full/ac7e046/ : SUCCESS in 56m 14s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/16/check/gate-grenade-dsvm/4347584/ : SUCCESS in 53m 45s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-large-ops/44d185a/ : SUCCESS in 37m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-neutron-large-ops/97f2d7a/ : SUCCESS in 37m 18s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/16/check/gate-nova-tox-functional/d4ec33d/ : SUCCESS in 17m 16s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/16/check/gate-grenade-dsvm-multinode/c17699f/ : SUCCESS in 1h 01m 27s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/87e9d1f/ : SUCCESS in 50m 21s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-nova-v20-api/2ab5ce6/ : SUCCESS in 45m 31s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-nova-v20-api-legacy/df8963e/ : SUCCESS in 38m 07s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-cells/c76dd0a/ : SUCCESS in 39m 00s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-full-ceph/f4c0164/ : SUCCESS in 46m 05s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/16/check/gate-tempest-dsvm-multinode-full/cf2bf25/ : FAILURE in 1h 08m 46s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/16/check/gate-compute-api-guide/f1004eb//api-guide/build/html/ : SUCCESS in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d6df953343e7ebb21c71a78bd807e926b20c0909","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-16 11:11:41.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/16/ : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":16},{"id":"fc2f7e71d0a96c66eae428e046e160eb3f4efb98","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-16 11:12:33.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/16/check/dsvm-tempest-xen/b72153d : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":16},{"id":"825998eb3d70410247fb9edaa8de1bc906b99cd2","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-11-16 12:46:42.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/80/175780/16/check/dsvm-tempest-cinder/bab6e38/ : SUCCESS in 54m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"33774565d932c7a1ecdafc9084f1a617c8aba966","author":{"_account_id":15424,"name":"javeme","email":"javaloveme@gmail.com","username":"javeme"},"date":"2015-11-16 12:53:34.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"98d9e19518a3decb0b348e01b0ac6850f4067883","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-11-16 13:51:32.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/80/175780/16/check/check-dsvm-vz7-tempest-exe-minimal/e78cfe9/ : FAILURE in 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"ba310bab79c5dde93a43ae85525918ffbcdbd22c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-11-16 15:33:50.000000000","message":"Patch Set 16:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/80/175780/16/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":16},{"id":"651c69e22ab8a5dfc98fcea5b813e3449bb503eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-16 22:40:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/16 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"e6d86975e287cc506f1b023510fbb2660188e169","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-19 15:04:33.000000000","message":"Patch Set 16: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"fa4f2a2c0767f4371616ebc0b0a7566b63db1d52","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-11-30 12:09:01.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f6b607619133d036b9d4ca17bf1e37fed8110df6","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-30 13:07:18.000000000","message":"Patch Set 17: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/17/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/884d29a29aa1469587056b51d767191c/index.html : SUCCESS in 16m 32s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/17/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9038ddf8abd24ac2ae01b09dd5ec96d5/index.html : SUCCESS in 17m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/17/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/143351b10796423bb7ceae0364db1439/index.html : SUCCESS in 15m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/17/check/real-db-upgrade_nova_percona_user_001:th-percona/5176de724af945c8a349e544901745d7/index.html : SUCCESS in 19m 20s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://localhost : FAILURE during cleanup and log upload\nException: Authorization Failure. Authorization Failed: Unable to establish connection to https://identity.api.rackspacecloud.com/v2.0/tokens in 15m 52s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/17/check/real-db-upgrade_nova_percona_user_002:th-percona/a90150af24af4d16afe89cc58ead5aba/index.html : SUCCESS in 16m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":17},{"id":"cb0caba10c9803bac50e572de6d0853dcfb9d021","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-30 13:10:19.000000000","message":"Patch Set 17: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/17/17441/index.html : SUCCESS in 58 m 14 s","accounts_in_message":[],"_revision_number":17},{"id":"ed1362e11753957efc39664deccc53e3aee51bc1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-30 13:16:31.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/17/ : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":17},{"id":"cd843ab5d64b4b0f20bfe591e9d4b637a2d961b8","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-30 13:54:55.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/17/check/dsvm-tempest-xen/94f9b01 : SUCCESS in 1h 36m 02s","accounts_in_message":[],"_revision_number":17},{"id":"82272e54a59f8938f39b294e82874413fa3da0ab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-30 14:07:07.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/17/check/gate-nova-pep8/b333751/ : SUCCESS in 7m 16s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/17/check/gate-nova-docs/2a116c9//doc/build/html/ : SUCCESS in 14m 06s\n- gate-nova-python27 http://logs.openstack.org/80/175780/17/check/gate-nova-python27/aede64e/ : SUCCESS in 15m 44s\n- gate-nova-python34 http://logs.openstack.org/80/175780/17/check/gate-nova-python34/4e50d8c/ : SUCCESS in 14m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-full/afbd7c9/ : SUCCESS in 48m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-postgres-full/b0ee80c/ : SUCCESS in 50m 11s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-neutron-full/b19a02b/ : SUCCESS in 1h 22m 36s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/17/check/gate-grenade-dsvm/9ceb2cc/ : SUCCESS in 45m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-large-ops/807857c/ : SUCCESS in 30m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-neutron-large-ops/aa43358/ : SUCCESS in 37m 23s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/17/check/gate-nova-tox-functional/5e1c796/ : SUCCESS in 16m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/17/check/gate-grenade-dsvm-multinode/dc44a24/ : SUCCESS in 53m 57s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8351e0a/ : SUCCESS in 40m 38s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-nova-v20-api/9db9ec0/ : SUCCESS in 38m 21s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-nova-v20-api-legacy/0ef4cda/ : SUCCESS in 46m 13s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-cells/705b2c8/ : SUCCESS in 52m 01s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-full-ceph/6408343/ : SUCCESS in 55m 34s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/17/check/gate-tempest-dsvm-multinode-full/b9c7fa4/ : FAILURE in 55m 28s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/17/check/gate-compute-api-guide/086b66a//api-guide/build/html/ : SUCCESS in 9m 00s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/80/175780/17/check/gate-nova-python27-constraints/2d7189c/ : SUCCESS in 12m 55s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/80/175780/17/check/gate-nova-python34-constraints/c9ee61d/ : SUCCESS in 11m 09s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/80/175780/17/check/gate-nova-docs-constraints/9294251//doc/build/html/ : SUCCESS in 14m 45s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/80/175780/17/check/gate-nova-pep8-constraints/4535c6d/ : SUCCESS in 9m 17s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/80/175780/17/check/gate-nova-tox-constraints-functional-constraints/ddf2611/ : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ac5f1ecbb8f437f90177fad4c982cedc5e4d52a3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-11-30 14:53:04.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/80/175780/17/check/dsvm-tempest-cinder/b796fb9/ : SUCCESS in 44m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"b6ced54ff873755c41dbf95ed29b255548f6de40","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-30 17:23:47.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/pci/175780/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"9aef8c5eac008ee01d153687af925b9278f489d1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-30 19:06:11.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/17/Nova-ML2-Sriov : SUCCESS in 1h 07m 09s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":17},{"id":"61d46f9988d17e7b7c55670c9a738dc9a0182195","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-01 07:11:18.000000000","message":"Patch Set 17: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/17 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/175780/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"3aec9f2cf5b9d8d31b55266a55bc7abe0d2d02f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-10 00:43:19.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"46ae5c013360720bb7b2c683814a6cf7c4623f78","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-12-10 09:55:12.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"01960c4974f76ba6d9fdd2d2f7586613e5493ba4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-10 09:55:52.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/175780/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"e25ed328a1b8e669b25cb23e7ec99021f16738cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-10 10:43:41.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/18/check/check-ibm-tempest-dsvm-full/972727d/ : SUCCESS in 47m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2328c90684db355682b8e3fc317320ae29deedeb","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-10 11:11:58.000000000","message":"Patch Set 18: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/18/18944/index.html : SUCCESS in 54 m 16 s","accounts_in_message":[],"_revision_number":18},{"id":"480c5cc681927cb262bcf699cfaab127d144dfa9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-10 11:35:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/18/check/gate-nova-pep8/4822f94/ : SUCCESS in 6m 53s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/18/check/gate-nova-docs/524efea//doc/build/html/ : SUCCESS in 28m 58s\n- gate-nova-python27 http://logs.openstack.org/80/175780/18/check/gate-nova-python27/bdf71af/ : SUCCESS in 12m 42s\n- gate-nova-python34 http://logs.openstack.org/80/175780/18/check/gate-nova-python34/238a62c/ : SUCCESS in 10m 27s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-full/96cd823/ : SUCCESS in 46m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-postgres-full/cbaf89a/ : SUCCESS in 58m 19s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-neutron-full/110e7a2/ : SUCCESS in 1h 24m 59s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/18/check/gate-grenade-dsvm/4437e22/ : SUCCESS in 1h 00m 01s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-large-ops/b7958e2/ : SUCCESS in 29m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-neutron-large-ops/aa81923/ : SUCCESS in 33m 25s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/18/check/gate-nova-tox-functional/b095c7f/ : SUCCESS in 14m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/18/check/gate-grenade-dsvm-multinode/4dcb79e/ : SUCCESS in 47m 37s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6ddef72/ : SUCCESS in 52m 44s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-nova-v20-api/6f37dd7/ : SUCCESS in 43m 00s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-nova-v20-api-legacy/6172d70/ : SUCCESS in 45m 05s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-cells/8adbc69/ : SUCCESS in 47m 35s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-full-ceph/9909f98/ : SUCCESS in 54m 57s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/18/check/gate-tempest-dsvm-multinode-full/9fb15d5/ : SUCCESS in 54m 43s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/18/check/gate-compute-api-guide/040be85//api-guide/build/html/ : SUCCESS in 11m 48s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/80/175780/18/check/gate-nova-python27-constraints/144a39a/ : SUCCESS in 14m 33s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/80/175780/18/check/gate-nova-python34-constraints/9ba6b3d/ : SUCCESS in 10m 04s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/80/175780/18/check/gate-nova-docs-constraints/55bc75a//doc/build/html/ : SUCCESS in 14m 35s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/80/175780/18/check/gate-nova-pep8-constraints/ac2a8e5/ : SUCCESS in 12m 09s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/80/175780/18/check/gate-nova-tox-constraints-functional-constraints/8df56f8/ : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"049707ec622f99d9a112c93408f14035881f0661","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-10 11:45:24.000000000","message":"Patch Set 18: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/18/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4986b5de267846a5941e8b8637f18833/index.html : SUCCESS in 19m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/18/check/real-db-upgrade_nova_mysql_user_001:th-mysql/635c92071ad8478d9218bdc1c970d34e/index.html : SUCCESS in 16m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/18/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/44aee110eb95403ea295f1a4bed5d8f0/index.html : SUCCESS in 12m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/18/check/real-db-upgrade_nova_percona_user_001:th-percona/47a761a26bec4294a7e8d0363a7ea04c/index.html : SUCCESS in 15m 07s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/18/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2c7d847ce06b480eb9fdb14ea5f4eb1e/index.html : SUCCESS in 14m 54s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/18/check/real-db-upgrade_nova_percona_user_002:th-percona/1a601e9e47f14a3fb75c04004214d24f/index.html : SUCCESS in 16m 19s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":18},{"id":"588e8189d806863616f6fe884e8c03d8e25ed09f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-10 11:49:50.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/175780/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"05eb5e0afdd23cbeadaaf1c4dd3ce5c2f0158531","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-10 12:08:08.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/18/check/dsvm-tempest-xen/18169fb : SUCCESS in 1h 17m 31s","accounts_in_message":[],"_revision_number":18},{"id":"e7d23584ed63040cf2646f592937c800b7029edc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-10 12:51:52.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/18/ : FAILURE in 1h 51m 40s","accounts_in_message":[],"_revision_number":18},{"id":"6f274454ffad330ec5d6ee305fb1bc3b6b476187","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-10 17:27:22.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/80/175780/18/check/check-dsvm-vz7-tempest-exe-minimal/f47426e/ : SUCCESS in 47m 30s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"54af9383c5197d62813d4f016d0dddc23390d6d2","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-10 19:35:21.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/80/175780/18/check/dsvm-tempest-cinder/82b7210/ : SUCCESS in 54m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"3e7c349477156d1c32bc1bcbd6fdc4f928019702","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-10 22:56:30.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/18/Nova-ML2-Sriov : SUCCESS in 58m 37s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":18},{"id":"bfc2fdf872f5a60dd036f137410b935a9c7d2907","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-11 02:19:57.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/175780/18 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"5243d5b5a3be6d8a76548688f9e06bffee284b23","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-11 12:38:38.000000000","message":"Patch Set 18:\n\nBuild failed. Logs: http://208.91.1.172/logs/175780/18 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/175780/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"8cd602dc1b75d57c5394027c685f0046e6292f16","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-12-15 15:30:46.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"e16c08fdd371ab7663952bb0d9bcaaf28fe3d286","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-15 16:47:39.000000000","message":"Patch Set 19:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/19/check/check-ibm-tempest-dsvm-full/5ef9058/ : SUCCESS in 1h 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"30d729140ae5e199818e53e42219682f5d7f08d5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-15 17:14:56.000000000","message":"Patch Set 19: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/19/19756/index.html : SUCCESS in 01 h 01 m 17 s","accounts_in_message":[],"_revision_number":19},{"id":"1f8cc772fe09748f3a199087312da9a5cb5f0016","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 17:53:55.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/19/check/gate-nova-pep8/8eff657/ : SUCCESS in 8m 50s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/19/check/gate-nova-docs/93c7281//doc/build/html/ : SUCCESS in 11m 08s\n- gate-nova-python27 http://logs.openstack.org/80/175780/19/check/gate-nova-python27/f7dccf5/ : FAILURE in 14m 41s\n- gate-nova-python34 http://logs.openstack.org/80/175780/19/check/gate-nova-python34/8cc06f0/ : SUCCESS in 13m 50s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-full/0574df6/ : SUCCESS in 44m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-postgres-full/b566521/ : SUCCESS in 41m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-neutron-full/5d6740f/ : SUCCESS in 57m 44s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/19/check/gate-grenade-dsvm/a62a046/ : SUCCESS in 39m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-large-ops/d2d6ce7/ : SUCCESS in 29m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-neutron-large-ops/f1a4fc0/ : SUCCESS in 27m 59s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/19/check/gate-nova-tox-functional/b7044fa/ : SUCCESS in 16m 32s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/19/check/gate-grenade-dsvm-multinode/e532078/ : SUCCESS in 58m 28s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/dd57591/ : SUCCESS in 45m 22s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-nova-v20-api/d0658dd/ : SUCCESS in 34m 08s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-nova-v20-api-legacy/d3d64e7/ : SUCCESS in 37m 44s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-cells/021d0eb/ : SUCCESS in 43m 05s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-full-ceph/3921012/ : SUCCESS in 46m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/82a0b52/ : SUCCESS in 47m 06s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-multinode-full/015b3a4/ : FAILURE in 55m 00s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/19/check/gate-compute-api-guide/96ad9a8//api-guide/build/html/ : SUCCESS in 8m 46s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-python27-constraints/0401434/ : SUCCESS in 14m 41s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-python34-constraints/b2ac479/ : SUCCESS in 14m 50s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/80/175780/19/check/gate-nova-docs-constraints/5611c29//doc/build/html/ : SUCCESS in 14m 42s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-pep8-constraints/4bbcbd6/ : SUCCESS in 9m 11s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-tox-constraints-functional-constraints/4726a78/ : SUCCESS in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ecad1de4b804989443b15bdccb0d78a33445d253","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-12-15 17:54:08.000000000","message":"Patch Set 19:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27: https://bugs.launchpad.net/bugs/1526413\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":19},{"id":"d73998e069797afbe125483d8337088f60407b33","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 17:57:27.000000000","message":"Patch Set 19:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":19},{"id":"229e11eecdd19e7cf99fd168b47fc090ae694219","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-15 18:12:29.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/19/check/dsvm-tempest-xen/4ca0e25 : SUCCESS in 1h 11m 50s","accounts_in_message":[],"_revision_number":19},{"id":"55a45dee31989764bf1fe7f6ce52d0e8665740c1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-15 18:24:34.000000000","message":"Patch Set 19: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/19/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/164e1d63061a48cf97dd5bdb30fe5362/index.html : SUCCESS in 14m 17s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/19/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1f8f2ff7d7944143b447d351369ab879/index.html : SUCCESS in 17m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/19/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d3ae4f2d66c249d990e65bdd6d3f89c1/index.html : SUCCESS in 15m 51s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/19/check/real-db-upgrade_nova_percona_user_001:th-percona/1d785583b983493cba3e5be8b9994115/index.html : SUCCESS in 20m 31s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/19/check/real-db-upgrade_nova_mysql_user_002:th-mysql/eab3a47434c748739a443f5c57ae0954/index.html : SUCCESS in 16m 11s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/19/check/real-db-upgrade_nova_percona_user_002:th-percona/44da828762d543daab5f2f99d9a6a914/index.html : SUCCESS in 15m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":19},{"id":"120e92bb131551d8198aded71054ab383b20579c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 19:48:36.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/19/ : FAILURE in 1h 50m 39s","accounts_in_message":[],"_revision_number":19},{"id":"ef057585ad2b93f5ff574e08695ea0d7c6ab4115","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-16 00:02:20.000000000","message":"Patch Set 19:\n\n* pci-test http://52.27.155.124/pci/175780/19 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"c901aea29734a60dcfc05a1e7687333dacf07c32","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-12-16 08:56:53.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"32783e4cb8b4b60a022dc7f77ccc68ef7b63d570","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-16 09:46:29.000000000","message":"Patch Set 19:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/19/check/check-ibm-tempest-dsvm-full/f61993d/ : SUCCESS in 46m 51s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5db9325824b0a1a689e9fe6ab406fbf4c3bc31f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-16 10:28:12.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/19/check/gate-nova-pep8/9b6b35d/ : SUCCESS in 9m 27s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/19/check/gate-nova-docs/a486b72//doc/build/html/ : SUCCESS in 14m 54s\n- gate-nova-python27 http://logs.openstack.org/80/175780/19/check/gate-nova-python27/3648af9/ : FAILURE in 13m 59s\n- gate-nova-python34 http://logs.openstack.org/80/175780/19/check/gate-nova-python34/d0c408b/ : SUCCESS in 13m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-full/5d0c95f/ : SUCCESS in 55m 20s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-postgres-full/ec4aa61/ : SUCCESS in 41m 35s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-neutron-full/17c80b6/ : SUCCESS in 1h 21m 54s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/19/check/gate-grenade-dsvm/6a95042/ : SUCCESS in 54m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-large-ops/59fc7a6/ : SUCCESS in 29m 16s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-neutron-large-ops/0d26eab/ : SUCCESS in 33m 09s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/19/check/gate-nova-tox-functional/85dbc85/ : SUCCESS in 16m 26s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/19/check/gate-grenade-dsvm-multinode/085ee40/ : SUCCESS in 1h 02m 35s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6182776/ : SUCCESS in 43m 41s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-nova-v20-api/3d61e49/ : SUCCESS in 36m 13s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-nova-v20-api-legacy/4cab1d0/ : SUCCESS in 43m 03s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-cells/57544e3/ : SUCCESS in 44m 18s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-full-ceph/5a1671d/ : FAILURE in 48m 08s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/174ea1e/ : SUCCESS in 59m 01s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/19/check/gate-tempest-dsvm-multinode-full/cceb6af/ : FAILURE in 59m 18s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/19/check/gate-compute-api-guide/236b486//api-guide/build/html/ : SUCCESS in 8m 27s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-python27-constraints/f0df610/ : SUCCESS in 14m 55s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-python34-constraints/cb712e0/ : SUCCESS in 14m 23s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/80/175780/19/check/gate-nova-docs-constraints/666681b//doc/build/html/ : SUCCESS in 15m 27s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-pep8-constraints/e0ba0b5/ : SUCCESS in 9m 51s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/80/175780/19/check/gate-nova-tox-constraints-functional-constraints/c8bdb4a/ : SUCCESS in 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d37e331219ba99f0f50d18644ae513fc8e0f23f7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-16 10:44:20.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/19/ : SUCCESS in 1h 45m 29s","accounts_in_message":[],"_revision_number":19},{"id":"c4cfdb46482739f6dcd1d665442afaa0c3a55cdb","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2015-12-23 09:37:30.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"f014bfcd83fa79f317ff9a876821c95ab7eb76f9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-23 09:55:13.000000000","message":"Patch Set 20:\n\n* pci-test http://52.27.155.124/pci/175780/20 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"e02a1e46608ffad833a6e145b2777cb938888d13","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-23 10:01:17.000000000","message":"Patch Set 20: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/20/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9f81f000e76a405db44ecb597892dd46/index.html : SUCCESS in 14m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/20/check/real-db-upgrade_nova_mysql_user_001:th-mysql/81db2255fdc345948d5ab575f8a7f3f8/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/20/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f05abe9c5b6a41069bc675490b4379b0/index.html : SUCCESS in 14m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/20/check/real-db-upgrade_nova_percona_user_001:th-percona/37129f20346045a3bd78c7df612938cd/index.html : SUCCESS in 16m 52s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/20/check/real-db-upgrade_nova_mysql_user_002:th-mysql/60bab7e4b77e48f385f1f8b317ea807b/index.html : SUCCESS in 14m 58s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/20/check/real-db-upgrade_nova_percona_user_002:th-percona/72ec78070fae4822988c99b6686d07c5/index.html : SUCCESS in 14m 31s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":20},{"id":"15eaf720578842a525b948dfa6476aac82990b8f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-23 10:08:25.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/80/175780/20/check/dsvm-tempest-kvm/e48a45c : FAILURE in 26m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"070c2eae2d58a5b7db40583b41af8183fc79dae7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-23 10:23:42.000000000","message":"Patch Set 20:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/20/check/check-ibm-tempest-dsvm-full/56e1f6a/ : SUCCESS in 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"7b7d99299e8adc320e6866a3b0b2e326e12063b5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-23 10:47:46.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/20/ : FAILURE in 1h 09m 34s","accounts_in_message":[],"_revision_number":20},{"id":"2a819bd2046e1aba294c32efbeb1724882e443cf","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-23 10:48:30.000000000","message":"Patch Set 20: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/20/20813/index.html : SUCCESS in 01 h 00 m 55 s","accounts_in_message":[],"_revision_number":20},{"id":"30e0d19dc8a9a0ff939b15e966012d23f3701f56","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-23 10:48:40.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/20/check/gate-nova-pep8/529cdd7/ : SUCCESS in 9m 12s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/20/check/gate-nova-docs/f8b5093//doc/build/html/ : SUCCESS in 14m 49s\n- gate-nova-python27 http://logs.openstack.org/80/175780/20/check/gate-nova-python27/ab02ab4/ : SUCCESS in 11m 37s\n- gate-nova-python34 http://logs.openstack.org/80/175780/20/check/gate-nova-python34/4f59876/ : SUCCESS in 13m 21s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-full/c5f8633/ : SUCCESS in 40m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-postgres-full/afa0a5d/ : SUCCESS in 1h 01m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-neutron-full/de834a3/ : SUCCESS in 54m 14s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/20/check/gate-grenade-dsvm/b5448e4/ : SUCCESS in 45m 04s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-large-ops/0649286/ : SUCCESS in 31m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-neutron-large-ops/c542d48/ : SUCCESS in 30m 21s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/20/check/gate-nova-tox-functional/50e6223/ : SUCCESS in 17m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/20/check/gate-grenade-dsvm-multinode/edc637d/ : SUCCESS in 56m 11s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e608068/ : SUCCESS in 52m 03s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-nova-v20-api/8f8a133/ : SUCCESS in 40m 17s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-nova-v20-api-legacy/0d9eac2/ : SUCCESS in 36m 39s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-cells/8862519/ : SUCCESS in 1h 00m 24s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-full-ceph/d718ad1/ : SUCCESS in 1h 07m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/3600bd0/ : SUCCESS in 42m 11s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/20/check/gate-tempest-dsvm-multinode-full/16c4d38/ : FAILURE in 50m 50s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/20/check/gate-compute-api-guide/3ea2801//api-guide/build/html/ : SUCCESS in 7m 06s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/80/175780/20/check/gate-nova-commit-message/6ae8f1a/ : SUCCESS in 1m 37s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/80/175780/20/check/gate-nova-python27-constraints/46f9d0a/ : SUCCESS in 14m 40s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/80/175780/20/check/gate-nova-python34-constraints/f976e4a/ : SUCCESS in 14m 13s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/80/175780/20/check/gate-nova-docs-constraints/0b57efc//doc/build/html/ : SUCCESS in 15m 04s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/80/175780/20/check/gate-nova-pep8-constraints/7f689f6/ : SUCCESS in 10m 00s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/80/175780/20/check/gate-nova-tox-constraints-functional-constraints/6ca47b9/ : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4f08dfec527d142561419e035cfbb63c88007791","author":{"_account_id":15882,"name":"Victor Laza","email":"vlaza@cloudbasesolutions.com","username":"vlaza"},"date":"2015-12-23 11:21:43.000000000","message":"Patch Set 20:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":20},{"id":"3526c36b311d03a48840fa8344987a6f891cb873","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-23 12:25:59.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/20/ : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":20},{"id":"1bbd30b7004bfd6803aa46be2da47a0df710bc80","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-23 17:05:16.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/20/check/dsvm-tempest-xen/a82ee8e : SUCCESS in 1h 10m 20s","accounts_in_message":[],"_revision_number":20},{"id":"77826ae5da394e99644b9596a2b5d79779281ad6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-23 18:58:32.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/20/Nova-ML2-Sriov : SUCCESS in 56m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":20},{"id":"78075bea0b9daa6920133971797114d8bc207930","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-24 03:40:19.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/175780/20/check/check-dsvm-tempest-vz7-exe-minimal/014ebf7 : FAILURE in 3h 05m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"6d9f76d0a09eee3e40d2d0f767e21900e6f43d08","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-29 20:36:07.000000000","message":"Patch Set 20: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/175780/20 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/638/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"899927a62dbdaec1a1d004cda4b56e40b36ae6c2","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2016-01-11 17:32:49.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"9f2f93934bb037a758594dcf0019575ecf9c70d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-14 09:12:58.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"f12337895c1afb8ffd7a3b50b953593ab4491bb2","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-01-14 10:45:13.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"67c118bec8e8487871c9198b5530273c1890d3d5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-14 11:43:45.000000000","message":"Patch Set 21:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/21/check/check-ibm-tempest-dsvm-full/7ed35f4/ : SUCCESS in 57m 41s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"3de5a5870139985ac6ec81369e4804913c0c5e57","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-14 11:47:58.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/175780/21/check/check-dsvm-tempest-vz7-exe-minimal/fd0f743 : FAILURE in 5s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"d8d7b5f7839a13b75c2ef1720b08bce7abfdc2f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-14 11:50:34.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/21/check/gate-nova-pep8/088592c/ : SUCCESS in 8m 00s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/21/check/gate-nova-docs/c19f66f//doc/build/html/ : SUCCESS in 14m 44s\n- gate-nova-python27 http://logs.openstack.org/80/175780/21/check/gate-nova-python27/713c49d/ : SUCCESS in 11m 31s\n- gate-nova-python34 http://logs.openstack.org/80/175780/21/check/gate-nova-python34/5ad4c81/ : SUCCESS in 28m 56s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-full/eceb31c/ : SUCCESS in 46m 24s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-postgres-full/828b2ac/ : SUCCESS in 52m 32s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-neutron-full/0839b02/ : SUCCESS in 59m 15s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/21/check/gate-grenade-dsvm/d2714a9/ : FAILURE in 40m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-large-ops/50906b4/ : SUCCESS in 28m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-neutron-large-ops/44adffe/ : SUCCESS in 32m 28s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/21/check/gate-nova-tox-functional/09f6d2f/ : SUCCESS in 16m 32s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/21/check/gate-grenade-dsvm-multinode/05b8476/ : SUCCESS in 49m 50s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/dcf0cc9/ : SUCCESS in 45m 08s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-nova-v20-api/9dddd20/ : SUCCESS in 40m 10s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-nova-v20-api-legacy/43d0d1f/ : SUCCESS in 43m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-cells/b697b00/ : SUCCESS in 47m 53s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-full-ceph/6c62a26/ : SUCCESS in 55m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5b97287/ : SUCCESS in 53m 31s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/21/check/gate-tempest-dsvm-multinode-full/f713e24/ : FAILURE in 57m 08s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/80/175780/21/check/gate-compute-api-guide/6bb49b1//api-guide/build/html/ : SUCCESS in 10m 00s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/80/175780/21/check/gate-nova-commit-message/6454cb5/ : SUCCESS in 1m 36s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/80/175780/21/check/gate-nova-python27-constraints/e49b236/ : SUCCESS in 16m 59s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/80/175780/21/check/gate-nova-python34-constraints/30d1f5d/ : SUCCESS in 30m 06s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/80/175780/21/check/gate-nova-docs-constraints/cfff69d//doc/build/html/ : SUCCESS in 15m 16s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/80/175780/21/check/gate-nova-pep8-constraints/3551534/ : SUCCESS in 10m 00s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/80/175780/21/check/gate-nova-tox-constraints-functional-constraints/a767fca/ : SUCCESS in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a4fb8ff51b8b77c7705ed57bc0beed14d23f297a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-14 12:09:00.000000000","message":"Patch Set 21: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/21/23208/index.html : SUCCESS in 55 m 29 s","accounts_in_message":[],"_revision_number":21},{"id":"1b6bf1e3ce8d54d27919e3b7e8ddaac4ea11420d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-14 12:12:20.000000000","message":"Patch Set 21:\n\n* pci-test http://52.27.155.124/pci/175780/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"af9a8628530dd8efa86ffc87f996850eedf63355","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-14 12:37:48.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/21/check/dsvm-tempest-xen/2b4c0f5 : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":21},{"id":"5dd4483c9c1dc6acef4f04a744f2a8bc2eaba5db","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-14 12:43:58.000000000","message":"Patch Set 21: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/21/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a4d68cbddc9f4d0b939fc56f32fbb1c0/index.html : SUCCESS in 14m 45s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/21/check/real-db-upgrade_nova_mysql_user_001:th-mysql/72a3f94ec3854249854484d9731d0e2b/index.html : SUCCESS in 14m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/21/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ec2285d305684b32b8817a9de58c21e1/index.html : SUCCESS in 15m 19s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/21/check/real-db-upgrade_nova_percona_user_001:th-percona/186d22f098d244ed9f530321c69c6899/index.html : SUCCESS in 16m 24s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/21/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a91a8cabdb594ee28c14f7ce8cf7cc33/index.html : SUCCESS in 14m 22s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/21/check/real-db-upgrade_nova_percona_user_002:th-percona/9228daad68df4d5fbebaa3d122cccec3/index.html : SUCCESS in 15m 25s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":21},{"id":"60ee2eb07ec0596e570b90a6a8ccbbdf5455903d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-14 12:56:56.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/21/ : FAILURE in 1h 40m 45s","accounts_in_message":[],"_revision_number":21},{"id":"040d151afd2353b29b210497ad20c10e4f54ad37","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-17 02:38:42.000000000","message":"Patch Set 21: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/175780/21/1274 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"b4e53f1688ca0ad7da6d831edc70d13b18f2ddb2","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2016-01-19 10:57:42.000000000","message":"Patch Set 21: Code-Review-1\n\n(3 comments)\n\nLooks good, just a few things to sort out","accounts_in_message":[],"_revision_number":21},{"id":"9240e30eff6cb87fdbd5360f96100e628df24b06","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-01-27 10:45:24.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"22a728498d6382dfd044323f2dc5f101a3fdfcfc","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-01-27 12:06:10.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"fbf6fa93cd2ed83c6e42abdcd9fcf64436ccf05e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-27 12:25:50.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/175780/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"37dc21558f5c5030e61cfcf148f47fe21348b191","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-27 12:50:58.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/175780/22/check/check-dsvm-tempest-vz7-exe-minimal/24e19e5 : SUCCESS in 43m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"a0032c2671263947827d2b306120660c2d66b192","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-27 13:21:54.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/175780/22/check/dsvm-tempest-xen/bf56c79 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":22},{"id":"de74c95af67dbf1db05feb2591ff44d3bca904b7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-27 13:24:04.000000000","message":"Patch Set 22: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/22/4377/index.html : SUCCESS in 01 h 14 m 05 s","accounts_in_message":[],"_revision_number":22},{"id":"28480715ae584d4a2a5168e27edd1cf713baacde","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-27 13:25:28.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/22/check/gate-nova-pep8/6286cfd/ : SUCCESS in 7m 09s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/22/check/gate-nova-docs/2dc7ea3//doc/build/html/ : SUCCESS in 10m 49s\n- gate-nova-python27 http://logs.openstack.org/80/175780/22/check/gate-nova-python27/ca56b5f/ : SUCCESS in 14m 11s\n- gate-nova-python34 http://logs.openstack.org/80/175780/22/check/gate-nova-python34/210071c/ : SUCCESS in 27m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-full/e80443c/ : SUCCESS in 52m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-postgres-full/83ae2fa/ : SUCCESS in 42m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-neutron-full/4eb43cf/ : SUCCESS in 57m 43s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/22/check/gate-grenade-dsvm/5f6d2f1/ : SUCCESS in 54m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-large-ops/0e620d9/ : SUCCESS in 34m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-neutron-large-ops/88c7338/ : SUCCESS in 36m 40s\n- gate-nova-releasenotes http://docs-draft.openstack.org/80/175780/22/check/gate-nova-releasenotes/3398a02//releasenotes/build/html/ : SUCCESS in 7m 27s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/22/check/gate-nova-tox-functional/0529e23/ : SUCCESS in 16m 10s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/22/check/gate-grenade-dsvm-multinode/bfd7c09/ : FAILURE in 45m 25s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/d796da6/ : FAILURE in 40m 57s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-nova-v20-api/0ad7b58/ : SUCCESS in 40m 22s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-nova-v20-api-legacy/c05e500/ : SUCCESS in 35m 27s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-cells/0c09118/ : SUCCESS in 43m 51s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-full-ceph/60d95cc/ : SUCCESS in 49m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/11056a6/ : SUCCESS in 48m 53s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/22/check/gate-tempest-dsvm-multinode-full/6e69439/ : FAILURE in 1h 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"882a366b5d5a594055dbee5b09f64c6a2b0a7c3a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-27 14:47:18.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/22/ : FAILURE in 2h 39m 49s","accounts_in_message":[],"_revision_number":22},{"id":"efd8a0a9457a92a6aa2da9ec41dcd23c8479f4eb","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-27 15:55:58.000000000","message":"Patch Set 22:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/80/175780/22 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":22},{"id":"982d2caa7dc397fa1643b7f63a209ad38cb6e776","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-27 17:28:16.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/22/Nova-ML2-Sriov : SUCCESS in 1h 34m 51s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"c1b42eadb3cd5526b5dd526a266eb4bf26c991da","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-28 09:06:54.000000000","message":"Patch Set 22:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/175780/22/1790 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"957412e558a4f1b98db30e921240a9970b824ed9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-29 14:37:29.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/175780/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"5eb0d9e163385439c431de0a3b638e604e8fe194","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-30 09:43:38.000000000","message":"Patch Set 22: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/22/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/74e133d0772542f0af0443e962c98bfd/index.html : SUCCESS in 11m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/22/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f86341c7867e4fd4b55e0f2f7c2f544c/index.html : SUCCESS in 15m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/22/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/49bd6758045e458b9055b647656af0e4/index.html : SUCCESS in 14m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/22/check/real-db-upgrade_nova_percona_user_001:th-percona/5a7b16f61a6c4ac18a1176dd18d962fb/index.html : SUCCESS in 13m 06s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/22/check/real-db-upgrade_nova_mysql_user_002:th-mysql/582210ff036d449e853384cf875d7029/index.html : SUCCESS in 14m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/22/check/real-db-upgrade_nova_percona_user_002:th-percona/ba075dbcdf204ee6926c412c108719c0/index.html : SUCCESS in 16m 17s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":22},{"id":"7d54e77046230ba9f97fecec2d8c63824c191d74","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 23:19:28.000000000","message":"Patch Set 22: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":22},{"id":"575dc69fe814b0097f8636f3f2157f66706525b3","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-02-07 12:59:27.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"426b33e586470c083648bfd9eb342a83ecc033ad","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-07 13:16:35.000000000","message":"Patch Set 23: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/23/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2f30e8e437784bca8aada75b462a2ee6/index.html : SUCCESS in 14m 02s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/23/check/real-db-upgrade_nova_mysql_user_001:th-mysql/34dbea126c0249e3a2c56aee4847e8e2/index.html : SUCCESS in 15m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/23/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/19513c14392e400e8f636675782a6190/index.html : SUCCESS in 13m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/23/check/real-db-upgrade_nova_percona_user_001:th-percona/db0669a4ce78408c974ff0e67e648e1c/index.html : SUCCESS in 15m 51s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/23/check/real-db-upgrade_nova_mysql_user_002:th-mysql/04cc4bed6b484380a3e4f5b77ba508ee/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/23/check/real-db-upgrade_nova_percona_user_002:th-percona/44dae7b726bd4f369fab7a94401b3716/index.html : SUCCESS in 12m 04s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":23},{"id":"b6fe3e44e3f3e723138d9a913ad07efffbbb5bf3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-07 13:17:15.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/pci/175780/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"981f05d94f342f4f4584333cf4dcb3437804398d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-07 13:53:01.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/175780/23/check/check-dsvm-tempest-vz7-exe-minimal/a266dbd : SUCCESS in 52m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"5454ba469d19a59c42af0225a6d9600ef4cc8211","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-07 13:57:35.000000000","message":"Patch Set 23:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/23/check/check-ibm-tempest-dsvm-full/aaf4db3/ : SUCCESS in 57m 33s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"8793ab1776d5bf6cd50f326f598aaa0bc96789a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-07 13:58:09.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/23/check/gate-nova-pep8/c545960/ : SUCCESS in 3m 25s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/23/check/gate-nova-docs/eaccb1a//doc/build/html/ : SUCCESS in 7m 45s\n- gate-nova-python27 http://logs.openstack.org/80/175780/23/check/gate-nova-python27/0be72cb/ : SUCCESS in 11m 06s\n- gate-nova-python34 http://logs.openstack.org/80/175780/23/check/gate-nova-python34/f7f0b7c/ : SUCCESS in 29m 25s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-full/e54151a/ : SUCCESS in 42m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-postgres-full/ec1c79f/ : SUCCESS in 40m 11s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-neutron-full/773ddc7/ : SUCCESS in 49m 10s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/23/check/gate-grenade-dsvm/a414f54/ : SUCCESS in 37m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-large-ops/47e9777/ : SUCCESS in 26m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-neutron-large-ops/cae8578/ : SUCCESS in 26m 13s\n- gate-nova-releasenotes http://docs-draft.openstack.org/80/175780/23/check/gate-nova-releasenotes/a1803f4//releasenotes/build/html/ : SUCCESS in 4m 12s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/23/check/gate-nova-tox-functional/d5d4b4e/ : SUCCESS in 13m 03s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/23/check/gate-grenade-dsvm-multinode/a5000f4/ : SUCCESS in 43m 35s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9aa0dad/ : SUCCESS in 33m 01s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-nova-v20-api/b967410/ : SUCCESS in 30m 01s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-nova-v20-api-legacy/bd78754/ : SUCCESS in 30m 22s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-cells/64c56b6/ : SUCCESS in 34m 43s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-full-ceph/20d6607/ : SUCCESS in 41m 47s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/c979cc7/ : FAILURE in 43m 47s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/23/check/gate-tempest-dsvm-multinode-full/ec55319/ : FAILURE in 55m 28s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"34e9a42b09fbb3e7bfb0a5ad58746c826cc378c1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-07 14:25:19.000000000","message":"Patch Set 23: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/23/5479/index.html : SUCCESS in 01 h 21 m 11 s","accounts_in_message":[],"_revision_number":23},{"id":"86cc4f4c5b07a0f40cd3decff1a07e4a0b2784dc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-07 14:32:25.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-175780-23 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":23},{"id":"99e6c441e68bd128076ae5923964f1908b63e1f2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-07 14:54:17.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/23/Nova-ML2-Sriov : SUCCESS in 39m 42s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":23},{"id":"45de1ea8e18b2e8adb7a1bee175fe5fa429c370c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-07 16:29:15.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/23/ : FAILURE in 3h 29m 10s","accounts_in_message":[],"_revision_number":23},{"id":"6f696a4ea326d1b3b5ac99841155a2f92c2dd456","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-07 21:07:01.000000000","message":"Patch Set 23:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":23},{"id":"5ad8540da7688b264bb5266bdb162d396b9f01d5","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2016-02-07 21:43:29.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"0a4d61982023d1eb6802bcc9c8937ac4f0653d38","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-07 22:33:38.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/23/ : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":23},{"id":"6add33bd1762cdd2e667c8ee08dd760407dab3b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-11 07:02:46.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"d165a22c5c41af2e8f8243e46ddebcf4108d7078","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-02-11 12:03:48.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"1f1932c5dbd014080fbb455ba5944df654ab268b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-11 12:14:54.000000000","message":"Patch Set 24:\n\n* pci-test http://52.27.155.124/pci/175780/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"d9444e0a1b01f6ae71c5956d11c4d8f85a10aca7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-11 12:35:58.000000000","message":"Patch Set 24:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/175780/24/check/check-ibm-tempest-dsvm-full/0197014/ : FAILURE in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"a1e99504a1c47cac390936c515975175ae38cbf0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-11 12:49:09.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/175780/24/check/check-dsvm-tempest-vz7-exe-minimal/73c7822 : FAILURE in 10m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"f9933411e62fedd9508bef0365083cbe8857e79a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-11 12:51:59.000000000","message":"Patch Set 24: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/24/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a6d795479855490ba6b3893a2bd8d128/index.html : SUCCESS in 14m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/24/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3042b3905c4a4a93bd29e70a0dff1a94/index.html : SUCCESS in 13m 53s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/24/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c16cd5d47b5645ba8e77a3619b50ac8e/index.html : SUCCESS in 12m 04s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/24/check/real-db-upgrade_nova_percona_user_001:th-percona/7fa162a1f6a5462d8a0d38bca89ac22e/index.html : SUCCESS in 21m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/24/check/real-db-upgrade_nova_mysql_user_002:th-mysql/52c7e373111149a3b4fc52cbcac34593/index.html : SUCCESS in 14m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/24/check/real-db-upgrade_nova_percona_user_002:th-percona/794df970623b4b3b8e5de914d2bb0ee1/index.html : SUCCESS in 14m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":24},{"id":"b9e959678337e2157cfd52c61780c4b786fc18ee","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-11 13:22:42.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/24/check/gate-nova-pep8/dcbbbed/ : FAILURE in 3m 09s\n- gate-nova-docs http://logs.openstack.org/80/175780/24/check/gate-nova-docs/f314c48/ : FAILURE in 3m 01s\n- gate-nova-python27 http://logs.openstack.org/80/175780/24/check/gate-nova-python27/7e91b0c/ : FAILURE in 2m 28s\n- gate-nova-python34 http://logs.openstack.org/80/175780/24/check/gate-nova-python34/ec377ac/ : FAILURE in 8m 53s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-full/1d55e37/ : FAILURE in 26m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-postgres-full/1609761/ : FAILURE in 24m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-neutron-full/b1d52d9/ : FAILURE in 36m 55s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/24/check/gate-grenade-dsvm/797fe52/ : FAILURE in 36m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-large-ops/a172bbc/ : SUCCESS in 29m 35s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-neutron-large-ops/18d21a1/ : SUCCESS in 26m 44s\n- gate-nova-releasenotes http://docs-draft.openstack.org/80/175780/24/check/gate-nova-releasenotes/e48bcc0//releasenotes/build/html/ : SUCCESS in 2m 06s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/24/check/gate-nova-tox-functional/182fe11/ : FAILURE in 2m 16s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/24/check/gate-grenade-dsvm-multinode/2d8c012/ : FAILURE in 37m 26s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/395f8b6/ : SUCCESS in 36m 23s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-cells/504361b/ : FAILURE in 23m 24s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-full-ceph/fb6a8e5/ : FAILURE in 22m 56s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/99b8c56/ : FAILURE in 23m 05s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/24/check/gate-tempest-dsvm-multinode-full/4634356/ : FAILURE in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c4d662fadb82948edf225183cb1e0908b5ebefef","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-11 13:34:40.000000000","message":"Patch Set 24: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/24/6023/index.html : SUCCESS in 01 h 24 m 13 s","accounts_in_message":[],"_revision_number":24},{"id":"5de952c5dcb2a42d275815eff4ecc3df1c96978d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-11 14:45:49.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/24/ : SUCCESS in 2h 07m 50s","accounts_in_message":[],"_revision_number":24},{"id":"e6048632bf8c99c8aafd9b81d48c449c2affb12f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-11 21:11:37.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/24/Nova-ML2-Sriov : FAILURE in 24m 58s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":24},{"id":"9227413aa0bd3447ebf2dd82843e36da7443d1a3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 04:36:54.000000000","message":"Patch Set 24:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/80/175780/24 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":24},{"id":"3b0d8248e5d7c294d30ea5bb599ecdb29bd6315c","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-02-12 10:04:12.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"735b67481e7d8f10c6ebb65965d9835c22c857d1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-12 10:22:12.000000000","message":"Patch Set 25: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/25/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/21a213c2355b481498d07447f66d1f77/index.html : SUCCESS in 12m 10s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/25/check/real-db-upgrade_nova_mysql_user_001:th-mysql/49a1718ef14e48d28a50c3ba3159fc75/index.html : SUCCESS in 16m 41s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/25/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b45291ce10274c18ae3e18713fbded84/index.html : SUCCESS in 13m 56s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/25/check/real-db-upgrade_nova_percona_user_001:th-percona/828746d30eca4383af5d4a93fafed92b/index.html : SUCCESS in 14m 33s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/25/check/real-db-upgrade_nova_mysql_user_002:th-mysql/35f1b7d85ceb4095a670287c5b2069c1/index.html : SUCCESS in 14m 07s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/25/check/real-db-upgrade_nova_percona_user_002:th-percona/125b3c0c941149b593f72dfdaabb2799/index.html : SUCCESS in 12m 27s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":25},{"id":"def28877b46c1badabdc20973544377d12b469b7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-12 10:22:25.000000000","message":"Patch Set 25:\n\n* pci-test http://52.27.155.124/pci/175780/25 : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"7262187df0d8b49b40a387f41538b88478139a91","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 11:08:57.000000000","message":"Patch Set 25:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/175780/25/check/check-dsvm-tempest-vz7-exe-minimal/57c48a0 : FAILURE in 1h 02m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"05ff0d55649117daadec8e9d9b2dca5ce3fa605e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 11:20:15.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/25/check/gate-nova-pep8/ab0888b/ : SUCCESS in 2m 30s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/25/check/gate-nova-docs/ffa247a//doc/build/html/ : SUCCESS in 7m 03s\n- gate-nova-python27 http://logs.openstack.org/80/175780/25/check/gate-nova-python27/c4776ff/ : FAILURE in 9m 31s\n- gate-nova-python34 http://logs.openstack.org/80/175780/25/check/gate-nova-python34/41a9655/ : SUCCESS in 27m 31s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-full/9eb5a03/ : SUCCESS in 33m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-postgres-full/edab56e/ : SUCCESS in 37m 39s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-neutron-full/b9ab097/ : SUCCESS in 46m 50s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/25/check/gate-grenade-dsvm/651017e/ : SUCCESS in 50m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-large-ops/2496a5c/ : SUCCESS in 49m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-neutron-large-ops/03a2b01/ : SUCCESS in 22m 27s\n- gate-nova-releasenotes http://docs-draft.openstack.org/80/175780/25/check/gate-nova-releasenotes/ad3b247//releasenotes/build/html/ : SUCCESS in 2m 12s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/25/check/gate-nova-tox-functional/afa5ee0/ : SUCCESS in 11m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/25/check/gate-grenade-dsvm-multinode/04db684/ : SUCCESS in 40m 59s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f45d358/ : SUCCESS in 33m 17s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-cells/0e8beae/ : SUCCESS in 35m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-full-ceph/6aadec2/ : SUCCESS in 35m 24s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/214d18b/ : SUCCESS in 39m 06s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/25/check/gate-tempest-dsvm-multinode-full/33c0faf/ : FAILURE in 53m 39s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"343effe480189dbe02f2960e5d9544490236ceaa","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-12 11:24:33.000000000","message":"Patch Set 25: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/25/6139/index.html : SUCCESS in 01 h 16 m 04 s","accounts_in_message":[],"_revision_number":25},{"id":"7deeac35f69f0cfb087d62680a7fa595b54d5737","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 11:43:46.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/25/ : SUCCESS in 1h 38m 37s","accounts_in_message":[],"_revision_number":25},{"id":"d96d995700fc02c654c0799b395c7c7c4fc2ce32","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-12 20:57:30.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/25/Nova-ML2-Sriov : SUCCESS in 1h 01m 22s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":25},{"id":"957eec8b74f5deac910551dbbce54de261a52801","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-02-15 08:50:17.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"9ff7beea73da0a79131ba9c9b4287e7d71c2ad21","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-15 09:04:40.000000000","message":"Patch Set 26:\n\n* pci-test http://52.27.155.124/pci/175780/26 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"91fb7acbc69a5bfc068d50de3bc6e363609c1edc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-15 09:07:21.000000000","message":"Patch Set 26: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/26/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/689108485bed40e485ce1f4af87e0e3c/index.html : SUCCESS in 14m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/26/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e7266bc3296f4317bf59e2615964e8fa/index.html : SUCCESS in 15m 52s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/26/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e0e211314e1a4e1795ba22822e90f181/index.html : SUCCESS in 14m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/26/check/real-db-upgrade_nova_percona_user_001:th-percona/3ede826debaa48809f316ee098b69c8b/index.html : SUCCESS in 15m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/80/175780/26/check/real-db-upgrade_nova_mysql_user_002:th-mysql/06af4d41f34f4057bfb2be8f9f1f1325/index.html : SUCCESS in 11m 45s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/80/175780/26/check/real-db-upgrade_nova_percona_user_002:th-percona/95cbb09787fe4f9e87590203a34561f9/index.html : SUCCESS in 14m 08s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":26},{"id":"ccdc3ac8c36b45d2b0b7599c3f814b02528ffb9a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-15 09:45:32.000000000","message":"Patch Set 26:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/80/175780/26 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":26},{"id":"c3833c3e02582e02c60f1aa257f61709e9fdf898","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-15 10:04:06.000000000","message":"Patch Set 26: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/175780/26/6273/index.html : SUCCESS in 01 h 09 m 48 s","accounts_in_message":[],"_revision_number":26},{"id":"acddf5af474272d33db827632d8ee1345212cf05","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-15 10:05:27.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/80/175780/26/check/gate-nova-pep8/2827c1b/ : SUCCESS in 2m 26s\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/26/check/gate-nova-docs/23f4030//doc/build/html/ : SUCCESS in 6m 33s\n- gate-nova-python27 http://logs.openstack.org/80/175780/26/check/gate-nova-python27/c168691/ : SUCCESS in 9m 17s\n- gate-nova-python34 http://logs.openstack.org/80/175780/26/check/gate-nova-python34/42bf43c/ : SUCCESS in 28m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-full/637b9d3/ : SUCCESS in 31m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-postgres-full/6c7d767/ : SUCCESS in 33m 07s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-neutron-full/97d5bd5/ : SUCCESS in 50m 19s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/26/check/gate-grenade-dsvm/d1e5b4d/ : SUCCESS in 31m 01s\n- gate-nova-releasenotes http://docs-draft.openstack.org/80/175780/26/check/gate-nova-releasenotes/9f3ec55//releasenotes/build/html/ : SUCCESS in 2m 14s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/26/check/gate-nova-tox-functional/cfafe44/ : SUCCESS in 10m 54s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/26/check/gate-grenade-dsvm-multinode/18aa8a7/ : SUCCESS in 40m 07s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a313216/ : FAILURE in 44m 27s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-cells/a75535d/ : SUCCESS in 28m 03s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-full-ceph/230a0cf/ : SUCCESS in 46m 00s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4ba0cb8/ : SUCCESS in 57m 52s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/80/175780/26/check/gate-tempest-dsvm-multinode-full/052ff39/ : FAILURE in 1h 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"76d4666dd92da2eacbdc9991beb256637d7151a7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-15 10:35:18.000000000","message":"Patch Set 26:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-175780-26 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":26},{"id":"e544830d26176cc70f472c7b7439f2501c06c198","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-15 10:44:34.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/175780/26/ : SUCCESS in 1h 53m 27s","accounts_in_message":[],"_revision_number":26},{"id":"d7336d07e0264cf759a49f4fd8949cdb8f8bd52c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-15 11:04:04.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/175780/26/Nova-ML2-Sriov : SUCCESS in 57m 53s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":26},{"id":"a5eb187a8860c884524f77221b9ce26d861152f3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2016-02-15 11:48:57.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"8ecb8c8e400256b568399464e4fb88eae2470676","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2016-02-15 11:50:01.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"97007688a04696a78913c0fc2b9462bdd5650c58","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-02-18 14:09:23.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"cd932770ba60d02b124d5b1afbf12e28187248ea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-18 14:09:32.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":26},{"id":"215c4ad46552e13dd5525d8851b915876c921f25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-18 15:09:17.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/80/175780/26/gate/gate-nova-docs/ee6feea//doc/build/html/ : SUCCESS in 8m 19s\n- gate-nova-pep8 http://logs.openstack.org/80/175780/26/gate/gate-nova-pep8/c9e059e/ : SUCCESS in 3m 58s\n- gate-nova-python27 http://logs.openstack.org/80/175780/26/gate/gate-nova-python27/9c1a605/ : SUCCESS in 9m 34s\n- gate-nova-python34 http://logs.openstack.org/80/175780/26/gate/gate-nova-python34/19cc5b0/ : SUCCESS in 29m 17s\n- gate-tempest-dsvm-full http://logs.openstack.org/80/175780/26/gate/gate-tempest-dsvm-full/87e7745/ : SUCCESS in 50m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/80/175780/26/gate/gate-tempest-dsvm-postgres-full/8e62bd8/ : SUCCESS in 54m 27s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/80/175780/26/gate/gate-tempest-dsvm-neutron-full/eacd555/ : SUCCESS in 59m 26s\n- gate-grenade-dsvm http://logs.openstack.org/80/175780/26/gate/gate-grenade-dsvm/d92a25d/ : SUCCESS in 52m 39s\n- gate-nova-releasenotes http://docs-draft.openstack.org/80/175780/26/gate/gate-nova-releasenotes/ef70485//releasenotes/build/html/ : SUCCESS in 2m 14s\n- gate-nova-tox-functional http://logs.openstack.org/80/175780/26/gate/gate-nova-tox-functional/ed993d3/ : SUCCESS in 12m 44s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/175780/26/gate/gate-grenade-dsvm-multinode/7d38489/ : SUCCESS in 44m 45s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/80/175780/26/gate/gate-tempest-dsvm-full-ceph/c51101f/ : SUCCESS in 42m 58s\n- gate-tempest-dsvm-cells http://logs.openstack.org/80/175780/26/gate/gate-tempest-dsvm-cells/e336a6e/ : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":26},{"id":"56ce85494c4098a5df229ecc79a803823405c2c1","date":"2016-02-18 15:09:22.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":26},{"id":"092be7c8967e1055e9c3fe14b638bc6b9219d1a8","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-02-19 12:36:05.000000000","message":"Patch Set 26:\n\nThanks John.\n\nI will try to cover your comments in another patch.","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"3159c8fd5bea80c820e58bd38d96f5f8fe8f4503","revisions":{"ec04702b5718076a8d13f2e440aa91576053a77d":{"kind":"REWORK","_number":1,"created":"2015-04-21 09:05:11.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/1"}}},"commit":{"parents":[{"commit":"b8947eef39272fee939818f383dd21be612e97cc","subject":"Merge \"Spelling mistakes in nova/compute/api.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8947eef39272fee939818f383dd21be612e97cc"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:44:01.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependant on virt_type","message":"libvirt: make live_migration_uri flag dependant on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default valur of the \"live_migration_uri\" flag,\nthat now is dependant on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec04702b5718076a8d13f2e440aa91576053a77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec04702b5718076a8d13f2e440aa91576053a77d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b8947eef39272fee939818f383dd21be612e97cc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"16cd530fc0f74c19ba25b606b13427fb0b572d3b":{"kind":"REWORK","_number":2,"created":"2015-06-04 10:30:04.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/2"}}},"commit":{"parents":[{"commit":"33fca195b1f6bb8d7b1d402a12fce839d51afeb4","subject":"Merge \"Tolerate iso style timestamps for cells rpc communication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33fca195b1f6bb8d7b1d402a12fce839d51afeb4"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-06-04 09:26:20.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependant on virt_type","message":"libvirt: make live_migration_uri flag dependant on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default valur of the \"live_migration_uri\" flag,\nthat now is dependant on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16cd530fc0f74c19ba25b606b13427fb0b572d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16cd530fc0f74c19ba25b606b13427fb0b572d3b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"33fca195b1f6bb8d7b1d402a12fce839d51afeb4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f3fc851f689defbe059cb2370b335da1cbdfac12":{"kind":"REWORK","_number":3,"created":"2015-06-04 11:21:24.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/3"}}},"commit":{"parents":[{"commit":"33fca195b1f6bb8d7b1d402a12fce839d51afeb4","subject":"Merge \"Tolerate iso style timestamps for cells rpc communication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33fca195b1f6bb8d7b1d402a12fce839d51afeb4"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-06-04 11:20:16.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default valur of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3fc851f689defbe059cb2370b335da1cbdfac12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3fc851f689defbe059cb2370b335da1cbdfac12"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"33fca195b1f6bb8d7b1d402a12fce839d51afeb4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"10defb1f59ec29bedb5958162d8211b0f82ca885":{"kind":"REWORK","_number":4,"created":"2015-06-04 14:57:59.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/4"}}},"commit":{"parents":[{"commit":"33fca195b1f6bb8d7b1d402a12fce839d51afeb4","subject":"Merge \"Tolerate iso style timestamps for cells rpc communication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33fca195b1f6bb8d7b1d402a12fce839d51afeb4"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-06-04 14:57:44.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10defb1f59ec29bedb5958162d8211b0f82ca885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10defb1f59ec29bedb5958162d8211b0f82ca885"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"33fca195b1f6bb8d7b1d402a12fce839d51afeb4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c1d9ba14a37744aa47de38f4616563f886d2f4e5":{"kind":"REWORK","_number":5,"created":"2015-06-19 08:50:57.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/5"}}},"commit":{"parents":[{"commit":"7ac22585a37601e73af4f5c1473a4f02377ba5cc","subject":"Merge \"Update version for Liberty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ac22585a37601e73af4f5c1473a4f02377ba5cc"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-06-19 08:42:20.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1d9ba14a37744aa47de38f4616563f886d2f4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1d9ba14a37744aa47de38f4616563f886d2f4e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ac22585a37601e73af4f5c1473a4f02377ba5cc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c1aae14834582e656b020134b895b6bcc7efd0e0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2015-06-29 11:47:31.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/6"}}},"commit":{"parents":[{"commit":"6969f270c5035325c603ce7a98b1647b72bf5eaa","subject":"Merge \"Fix typos detected by toolkit misspellings.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6969f270c5035325c603ce7a98b1647b72bf5eaa"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-06-29 11:47:24.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1aae14834582e656b020134b895b6bcc7efd0e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1aae14834582e656b020134b895b6bcc7efd0e0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6969f270c5035325c603ce7a98b1647b72bf5eaa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2aeeadfe2fd1cb819a61adc1da6d1dc01f313d81":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2015-07-06 08:05:37.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/7"}}},"commit":{"parents":[{"commit":"74197aaa3ffe2e0068b20b2b67bd759431bd5343","subject":"Merge \"Remove db layer hard-code permission checks for network_disassociate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74197aaa3ffe2e0068b20b2b67bd759431bd5343"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-07-06 08:05:36.000000000","tz":0},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2aeeadfe2fd1cb819a61adc1da6d1dc01f313d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2aeeadfe2fd1cb819a61adc1da6d1dc01f313d81"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"74197aaa3ffe2e0068b20b2b67bd759431bd5343","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"16f624fd937b157eda8e3f3e6e310620f4d7152c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2015-07-27 09:18:02.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/8"}}},"commit":{"parents":[{"commit":"183cd88cb2f9781b53f71b6b161df401c286c9ff","subject":"Merge \"remove _rescan_iscsi from disconnect_volume_multipath_iscsi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/183cd88cb2f9781b53f71b6b161df401c286c9ff"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-07-27 09:18:01.000000000","tz":0},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16f624fd937b157eda8e3f3e6e310620f4d7152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16f624fd937b157eda8e3f3e6e310620f4d7152c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"183cd88cb2f9781b53f71b6b161df401c286c9ff","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"62655a831d5f1d9e10167500538c78ab4cdbd454":{"kind":"REWORK","_number":9,"created":"2015-08-31 08:54:50.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/9"}}},"commit":{"parents":[{"commit":"858114bb6b30c872a59bbc6da2bddb56c40af354","subject":"Merge \"Strip the extra properties out when using legacy v2 compatible middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/858114bb6b30c872a59bbc6da2bddb56c40af354"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-08-31 08:53:50.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62655a831d5f1d9e10167500538c78ab4cdbd454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62655a831d5f1d9e10167500538c78ab4cdbd454"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"858114bb6b30c872a59bbc6da2bddb56c40af354","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9b38044065114e4c0ecbd6876809f55658394479":{"kind":"REWORK","_number":10,"created":"2015-08-31 10:39:07.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/10"}}},"commit":{"parents":[{"commit":"858114bb6b30c872a59bbc6da2bddb56c40af354","subject":"Merge \"Strip the extra properties out when using legacy v2 compatible middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/858114bb6b30c872a59bbc6da2bddb56c40af354"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-08-31 10:38:55.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b38044065114e4c0ecbd6876809f55658394479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b38044065114e4c0ecbd6876809f55658394479"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"858114bb6b30c872a59bbc6da2bddb56c40af354","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"147850b4db5e81f9773f262dbb21b4bef289a9e2":{"kind":"REWORK","_number":11,"created":"2015-09-04 08:46:07.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/11"}}},"commit":{"parents":[{"commit":"bf0f4b10cd52c0fd7e8dec1c0abedf32fb1a9c33","subject":"Merge \"[libvirt] Move cleanup of imported files to imagebackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf0f4b10cd52c0fd7e8dec1c0abedf32fb1a9c33"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-09-04 08:45:33.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/147850b4db5e81f9773f262dbb21b4bef289a9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/147850b4db5e81f9773f262dbb21b4bef289a9e2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bf0f4b10cd52c0fd7e8dec1c0abedf32fb1a9c33","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9896b73006e97868b60bd5fefa50a5c68250a13d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2015-09-18 08:30:24.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/12"}}},"commit":{"parents":[{"commit":"40015597f1cc2b296d416a0a77c22e7f6230a626","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40015597f1cc2b296d416a0a77c22e7f6230a626"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-09-18 08:29:55.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9896b73006e97868b60bd5fefa50a5c68250a13d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9896b73006e97868b60bd5fefa50a5c68250a13d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40015597f1cc2b296d416a0a77c22e7f6230a626","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"30f2f04b282e07067f0bec95614c25c22621b866":{"kind":"REWORK","_number":13,"created":"2015-09-22 10:19:38.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/13"}}},"commit":{"parents":[{"commit":"666c743fdb5f723d13e97303279d6f7662110039","subject":"Merge \"live-migration: Logs exception if operation failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/666c743fdb5f723d13e97303279d6f7662110039"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-09-22 08:48:26.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30f2f04b282e07067f0bec95614c25c22621b866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30f2f04b282e07067f0bec95614c25c22621b866"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"666c743fdb5f723d13e97303279d6f7662110039","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"75052a08052c6ac71205c5d0e04c97cc18aba8c3":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2015-10-06 15:08:32.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/14"}}},"commit":{"parents":[{"commit":"f00d76b1111e42efbb120b870022128eee0fce83","subject":"Merge \"If rescue failed set instance to ERROR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f00d76b1111e42efbb120b870022128eee0fce83"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-10-06 15:05:58.000000000","tz":120},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75052a08052c6ac71205c5d0e04c97cc18aba8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75052a08052c6ac71205c5d0e04c97cc18aba8c3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f00d76b1111e42efbb120b870022128eee0fce83","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8de3545e6b4c14dfeef6af938bc44d570fa4ca13":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2015-11-04 08:28:17.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/15"}}},"commit":{"parents":[{"commit":"58f5eb5875ba5624f725560128300596eef75e15","subject":"Merge \"Add debug logging for when boot sequence is invalid in _validate_bdm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/58f5eb5875ba5624f725560128300596eef75e15"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-11-04 08:28:07.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8de3545e6b4c14dfeef6af938bc44d570fa4ca13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8de3545e6b4c14dfeef6af938bc44d570fa4ca13"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"58f5eb5875ba5624f725560128300596eef75e15","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bb8fa060e98605e057c1ce0bbc1842cbccb2db10":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2015-11-16 09:59:35.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/16"}}},"commit":{"parents":[{"commit":"876679c08f05eaa2a90b1f053b65b78a0af433df","subject":"cells is a sad panda about scheduler hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/876679c08f05eaa2a90b1f053b65b78a0af433df"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-11-16 09:38:24.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb8fa060e98605e057c1ce0bbc1842cbccb2db10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb8fa060e98605e057c1ce0bbc1842cbccb2db10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"876679c08f05eaa2a90b1f053b65b78a0af433df","is_merged_in_target_branch":true,"change_id":"Icc71e36f4ecb015dff9e806caacd31262f7e17f7","change_number":245382,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"73fac19b364e23bdd6f5ff849d724973dfd7801a":{"kind":"REWORK","_number":17,"created":"2015-11-30 12:09:01.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/17"}}},"commit":{"parents":[{"commit":"1427b9f5b67f8c46574450b9f37298222f9942e0","subject":"Merge \"docs: reorder move servers text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1427b9f5b67f8c46574450b9f37298222f9942e0"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-11-30 12:08:48.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/73fac19b364e23bdd6f5ff849d724973dfd7801a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/73fac19b364e23bdd6f5ff849d724973dfd7801a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1427b9f5b67f8c46574450b9f37298222f9942e0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fd621c0ce3b65eda454b5fd3aa41c13d2572c64e":{"kind":"REWORK","_number":18,"created":"2015-12-10 09:55:12.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/18"}}},"commit":{"parents":[{"commit":"84952379e05307029ffdec97e069bde12f18e8c2","subject":"Merge \"Replaces izip_longest with six.moves.zip_longest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84952379e05307029ffdec97e069bde12f18e8c2"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-12-10 09:54:50.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd621c0ce3b65eda454b5fd3aa41c13d2572c64e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd621c0ce3b65eda454b5fd3aa41c13d2572c64e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"84952379e05307029ffdec97e069bde12f18e8c2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d796209b3c70606b337ef77579f9e1b66490017c":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2015-12-15 15:30:46.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/19"}}},"commit":{"parents":[{"commit":"aeaf5bea81f96b1180dc24f1c9b393a6eccb69c7","subject":"Merge \"Add uuidsentinel test module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aeaf5bea81f96b1180dc24f1c9b393a6eccb69c7"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-12-15 15:30:44.000000000","tz":0},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d796209b3c70606b337ef77579f9e1b66490017c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d796209b3c70606b337ef77579f9e1b66490017c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aeaf5bea81f96b1180dc24f1c9b393a6eccb69c7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b49fe91bf51492b595db11a8d27ecb70c6b4fe15":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2015-12-23 09:37:30.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/20"}}},"commit":{"parents":[{"commit":"727656d2f59e26a8c86afdad9b3709ab8efde3c1","subject":"Merge \"Nuke EC2 API from api-paste and remove wsgi support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/727656d2f59e26a8c86afdad9b3709ab8efde3c1"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-12-23 09:37:29.000000000","tz":0},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b49fe91bf51492b595db11a8d27ecb70c6b4fe15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b49fe91bf51492b595db11a8d27ecb70c6b4fe15"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"727656d2f59e26a8c86afdad9b3709ab8efde3c1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"efc6cef6fdebcb7a365c002b4f228a7069c8712a":{"kind":"REWORK","_number":21,"created":"2016-01-14 10:45:13.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/21"}}},"commit":{"parents":[{"commit":"5d6e0086b578077e7479b78e282e3f7da7ffca71","subject":"Merge \"Actually pass the migration data object down to the virt drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d6e0086b578077e7479b78e282e3f7da7ffca71"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-01-14 10:20:57.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/efc6cef6fdebcb7a365c002b4f228a7069c8712a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/efc6cef6fdebcb7a365c002b4f228a7069c8712a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5d6e0086b578077e7479b78e282e3f7da7ffca71","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a5eb2025e5ce5b831dd2401d8e0f8a43512a0cc4":{"kind":"REWORK","_number":22,"created":"2016-01-27 12:06:10.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/22"}}},"commit":{"parents":[{"commit":"a37590c21bca175f642792b6707d2eb4e6349e51","subject":"Merge \"Enables conductor py3 unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a37590c21bca175f642792b6707d2eb4e6349e51"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-01-27 12:05:54.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a5eb2025e5ce5b831dd2401d8e0f8a43512a0cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a5eb2025e5ce5b831dd2401d8e0f8a43512a0cc4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a37590c21bca175f642792b6707d2eb4e6349e51","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"72a1c9d913c30203ae9f9ca3cbbf87326930a952":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2016-02-07 12:59:27.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/23"}}},"commit":{"parents":[{"commit":"197bd6dd1231f1f57cdd6c0acb1dfbdc3b2b0989","subject":"Merge \"libvirt: use osinfo when configuring the disk bus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/197bd6dd1231f1f57cdd6c0acb1dfbdc3b2b0989"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-02-07 12:55:18.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72a1c9d913c30203ae9f9ca3cbbf87326930a952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72a1c9d913c30203ae9f9ca3cbbf87326930a952"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"197bd6dd1231f1f57cdd6c0acb1dfbdc3b2b0989","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8cf3fed93560b09ed79820f285661d86c7c477ae":{"kind":"REWORK","_number":24,"created":"2016-02-11 12:03:48.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/24"}}},"commit":{"parents":[{"commit":"4c42235e1403cb8fce309efce2034d1962be995e","subject":"Merge \"Return HTTP 400 for invalid server-group uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c42235e1403cb8fce309efce2034d1962be995e"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-02-11 12:03:30.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8cf3fed93560b09ed79820f285661d86c7c477ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8cf3fed93560b09ed79820f285661d86c7c477ae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c42235e1403cb8fce309efce2034d1962be995e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a333ecbb7116181dd49423ceb811b99ef251e156":{"kind":"REWORK","_number":25,"created":"2016-02-12 10:04:12.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/25"}}},"commit":{"parents":[{"commit":"4c42235e1403cb8fce309efce2034d1962be995e","subject":"Merge \"Return HTTP 400 for invalid server-group uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c42235e1403cb8fce309efce2034d1962be995e"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-02-12 09:59:49.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a333ecbb7116181dd49423ceb811b99ef251e156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a333ecbb7116181dd49423ceb811b99ef251e156"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c42235e1403cb8fce309efce2034d1962be995e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3159c8fd5bea80c820e58bd38d96f5f8fe8f4503":{"kind":"REWORK","_number":26,"created":"2016-02-15 08:50:17.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/80/175780/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/175780/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/175780/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/175780/26"}}},"commit":{"parents":[{"commit":"4c42235e1403cb8fce309efce2034d1962be995e","subject":"Merge \"Return HTTP 400 for invalid server-group uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c42235e1403cb8fce309efce2034d1962be995e"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2015-04-21 08:39:30.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-02-15 08:50:01.000000000","tz":60},"subject":"libvirt: make live_migration_uri flag dependent on virt_type","message":"libvirt: make live_migration_uri flag dependent on virt_type\n\nThe default value for the \"live_migration_uri\" flag should be\ndependent on the \"virt_type\" flag, as the \"connection_uri\" flag is. This\nway, an operator can set the \"virt_type\" flag without the need to check\nfor each individual uri.\n\nDocImpact: Changed the default value of the \"live_migration_uri\" flag,\nthat now is dependent on the \"virt_type\".\nCloses-Bug: #1298242\nChange-Id: Id54f7bdfa14a19da41da554b13ba9496ee525c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3159c8fd5bea80c820e58bd38d96f5f8fe8f4503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3159c8fd5bea80c820e58bd38d96f5f8fe8f4503"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c42235e1403cb8fce309efce2034d1962be995e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
