)]}'
{"id":"openstack%2Fnova~175787","triplet_id":"openstack%2Fnova~stable%2Fkilo~If0c7ad20506c9dddf4dec1eb64c9d6dd4fb75633","project":"openstack/nova","branch":"stable/kilo","topic":"bug/1444021","hashtags":[],"change_id":"If0c7ad20506c9dddf4dec1eb64c9d6dd4fb75633","subject":"scheduler: re-calculate NUMA on consume_from_instance","status":"MERGED","created":"2015-04-21 09:40:14.000000000","updated":"2015-04-22 18:45:54.000000000","submitted":"2015-04-22 18:45:51.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c7a419cd0cdfd61200512c96f727f2116bde198","_number":175787,"virtual_id_number":175787,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2015-04-21 11:46:51.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-04-22 18:45:51.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2015-04-21 09:51:06.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-21 10:45:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2015-04-22 11:15:26.000000000","_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-04-22 08:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":1,"date":"2015-04-21 12:03:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-04-22 08:08:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-04-21 09:51:06.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 10:45:00.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 11:46:51.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-04-21 12:03:54.000000000","updated_by":{"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"},"reviewer":{"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"},"state":"REVIEWER"},{"updated":"2015-04-22 08:08:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2015-04-22 11:15:26.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"},{"updated":"2015-04-22 18:45:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4255a6e964796eec4cb29c41757602d2e56742e0","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-04-21 09:40:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c62cef26bded6662a69f75f5439c82c84741d1be","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-04-21 09:40:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.","accounts_in_message":[],"_revision_number":1},{"id":"ffe50a18961872da63af219258e23faab81e3489","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2015-04-21 09:50:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce3f50e4f5d3f546b37e39f534fbdfa3a6bb2e11","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-04-21 09:51:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.","accounts_in_message":[],"_revision_number":2},{"id":"3640333e6629c5e1f01ce621f9b6a4decf3a4541","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-04-21 10:45:00.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175787/2/check-ibm-tempest-dsvm-full/28cafad : SUCCESS in 53m 06s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/175787/2/check-ibm-tempest-dsvm-postgres-full/a40c764 : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4bbd09fdc1fba2d6ae99d339262da270fb1f9c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-21 11:11:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/87/175787/2/check/gate-nova-pep8/b570b17/ : SUCCESS in 8m 49s\n- gate-nova-docs http://docs-draft.openstack.org/87/175787/2/check/gate-nova-docs/7cb725a//doc/build/html/ : SUCCESS in 8m 39s\n- gate-nova-python27 http://logs.openstack.org/87/175787/2/check/gate-nova-python27/6d94089/ : SUCCESS in 13m 49s\n- check-tempest-dsvm-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full/ffd6b5f/ : SUCCESS in 45m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-postgres-full/3ef600b/ : SUCCESS in 43m 45s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-neutron-full/bc23258/ : SUCCESS in 1h 17m 51s\n- check-grenade-dsvm http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm/5b5ea32/ : SUCCESS in 39m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-large-ops/2233d33/ : SUCCESS in 21m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-neutron-large-ops/be92499/ : SUCCESS in 20m 43s\n- check-devstack-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-devstack-dsvm-cells/cb630ba/ : SUCCESS in 20m 10s\n- gate-nova-tox-functional http://logs.openstack.org/87/175787/2/check/gate-nova-tox-functional/7f73c86/ : SUCCESS in 11m 34s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm-partial-ncpu/01e549c/ : SUCCESS in 37m 49s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-ironic-pxe_ssh/5c7fa06/ : SUCCESS in 38m 29s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-nova-v21-full/6b90abc/ : SUCCESS in 49m 44s\n- check-tempest-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-cells/c0bc202/ : FAILURE in 39m 40s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full-ceph/1776a58/ : SUCCESS in 40m 08s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-multinode-full/444e4db/ : FAILURE in 52m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95e9e5ff5d66e91a6bf1e7207a5b0a516887b07b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-04-21 11:46:51.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/175787/2/ : FAILURE in 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"935246477b36dcd7767beb0a664396146c8f7ba9","author":{"_account_id":11647,"name":"Przemyslaw Czesnowicz","email":"przemyslaw.czesnowicz@intel.com","username":"pczesno"},"date":"2015-04-21 12:03:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"31642011fed3105fa10240dd3a7e1b528fef5be0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2015-04-22 08:08:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nIn RC2, merged in mater","accounts_in_message":[],"_revision_number":2},{"id":"d6a0d6548405578164896cb7cfa93c3482c4d45a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 08:08:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"7f5b7a4b3f7e7a2b54af2d65d8437ab15b11280d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 09:04:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/87/175787/2/gate/gate-nova-docs/5c729c5//doc/build/html/ : SUCCESS in 8m 36s\n- gate-nova-pep8 http://logs.openstack.org/87/175787/2/gate/gate-nova-pep8/ffdd4f1/ : SUCCESS in 8m 34s\n- gate-nova-python27 http://logs.openstack.org/87/175787/2/gate/gate-nova-python27/c9cded6/ : SUCCESS in 14m 20s\n- gate-tempest-dsvm-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-full/a936d51/ : SUCCESS in 39m 41s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-postgres-full/ef106f3/ : SUCCESS in 40m 16s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-neutron-full/09b8236/ : SUCCESS in 54m 20s\n- gate-grenade-dsvm http://logs.openstack.org/87/175787/2/gate/gate-grenade-dsvm/5e46ced/ : FAILURE in 17m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-large-ops/481edb8/ : SUCCESS in 23m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-neutron-large-ops/49d7ab8/ : SUCCESS in 22m 52s\n- gate-nova-tox-functional http://logs.openstack.org/87/175787/2/gate/gate-nova-tox-functional/1395cce/ : SUCCESS in 11m 15s\n- gate-devstack-dsvm-cells http://logs.openstack.org/87/175787/2/gate/gate-devstack-dsvm-cells/2f19908/ : SUCCESS in 21m 01s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/87/175787/2/gate/gate-grenade-dsvm-partial-ncpu/8b542c6/ : FAILURE in 16m 25s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-nova-v21-full/5394d1b/ : SUCCESS in 47m 14s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-full-ceph/1a1ccc7/ : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"07bce5207e9f9cb0168d87a5cdbabd9e955ff5b3","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2015-04-22 11:15:26.000000000","message":"Patch Set 2:\n\nreverify","accounts_in_message":[],"_revision_number":2},{"id":"6359141ada478d066e587f3c9d76cac88f66d9e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 12:14:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/87/175787/2/check/gate-nova-pep8/e1f0331/ : SUCCESS in 8m 18s\n- gate-nova-docs http://docs-draft.openstack.org/87/175787/2/check/gate-nova-docs/a695b63//doc/build/html/ : SUCCESS in 8m 17s\n- gate-nova-python27 http://logs.openstack.org/87/175787/2/check/gate-nova-python27/2a6d1ff/ : SUCCESS in 16m 54s\n- check-tempest-dsvm-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full/780fe85/ : SUCCESS in 39m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-postgres-full/3688d3f/ : SUCCESS in 41m 35s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-neutron-full/02966b3/ : SUCCESS in 55m 18s\n- check-grenade-dsvm http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm/919d4b4/ : FAILURE in 16m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-large-ops/c59c61a/ : SUCCESS in 23m 17s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-neutron-large-ops/2678664/ : SUCCESS in 20m 03s\n- check-devstack-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-devstack-dsvm-cells/07f83de/ : SUCCESS in 20m 40s\n- gate-nova-tox-functional http://logs.openstack.org/87/175787/2/check/gate-nova-tox-functional/67e5ddd/ : SUCCESS in 12m 16s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm-partial-ncpu/0687aed/ : FAILURE in 14m 40s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-ironic-pxe_ssh/a70fd71/ : SUCCESS in 32m 10s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-nova-v21-full/072a06f/ : SUCCESS in 40m 29s\n- check-tempest-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-cells/37ec5ab/ : FAILURE in 34m 04s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full-ceph/6b42364/ : SUCCESS in 40m 46s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-multinode-full/af1e1c7/ : SUCCESS in 57m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09eee4ab989b5bf3b88a2a35f9c064b1d716da3b","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2015-04-22 12:22:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0de5e246a4c5de7f209e34b08c39115e5764504f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 13:21:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/87/175787/2/check/gate-nova-pep8/66ab9b2/ : SUCCESS in 8m 27s\n- gate-nova-docs http://docs-draft.openstack.org/87/175787/2/check/gate-nova-docs/5bd9271//doc/build/html/ : SUCCESS in 8m 11s\n- gate-nova-python27 http://logs.openstack.org/87/175787/2/check/gate-nova-python27/e09bcdb/ : SUCCESS in 14m 47s\n- check-tempest-dsvm-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full/4e08383/ : SUCCESS in 41m 57s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-postgres-full/04c4ad4/ : SUCCESS in 43m 22s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-neutron-full/fa1f742/ : SUCCESS in 56m 53s\n- check-grenade-dsvm http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm/fe7007d/ : FAILURE in 15m 00s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-large-ops/ee15594/ : SUCCESS in 24m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-neutron-large-ops/eda2eb1/ : SUCCESS in 21m 05s\n- check-devstack-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-devstack-dsvm-cells/6fb0e9c/ : SUCCESS in 21m 53s\n- gate-nova-tox-functional http://logs.openstack.org/87/175787/2/check/gate-nova-tox-functional/1fcee96/ : SUCCESS in 12m 11s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm-partial-ncpu/00d2022/ : FAILURE in 17m 52s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-ironic-pxe_ssh/9837a06/ : SUCCESS in 32m 38s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-nova-v21-full/841526e/ : SUCCESS in 43m 22s\n- check-tempest-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-cells/fcb628f/ : FAILURE in 40m 23s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full-ceph/a96d367/ : SUCCESS in 43m 06s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-multinode-full/85dccd2/ : SUCCESS in 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df7eea72c24102561e60aae91f2b552b5961c95d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2015-04-22 14:13:13.000000000","message":"Patch Set 2:\n\nrecheck should be ok now","accounts_in_message":[],"_revision_number":2},{"id":"c1d0d2a3740cc4851640eb9f2bbae59bb233b43b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 16:00:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/87/175787/2/check/gate-nova-pep8/5ed782e/ : SUCCESS in 8m 05s\n- gate-nova-docs http://docs-draft.openstack.org/87/175787/2/check/gate-nova-docs/d1c0e9a//doc/build/html/ : SUCCESS in 8m 29s\n- gate-nova-python27 http://logs.openstack.org/87/175787/2/check/gate-nova-python27/3d1c352/ : SUCCESS in 15m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full/c005e0a/ : SUCCESS in 43m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-postgres-full/052cce1/ : SUCCESS in 48m 51s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-neutron-full/ca948e6/ : SUCCESS in 53m 24s\n- check-grenade-dsvm http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm/7001118/ : SUCCESS in 38m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-large-ops/4ef2ff7/ : SUCCESS in 21m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/87/175787/2/check/gate-tempest-dsvm-neutron-large-ops/ec5d485/ : SUCCESS in 23m 56s\n- check-devstack-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-devstack-dsvm-cells/9eefd3c/ : SUCCESS in 18m 32s\n- gate-nova-tox-functional http://logs.openstack.org/87/175787/2/check/gate-nova-tox-functional/9cf68b2/ : SUCCESS in 11m 08s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/87/175787/2/check/check-grenade-dsvm-partial-ncpu/32f2a43/ : SUCCESS in 37m 41s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-ironic-pxe_ssh/bd67c1e/ : SUCCESS in 34m 23s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-nova-v21-full/76545e0/ : SUCCESS in 44m 35s\n- check-tempest-dsvm-cells http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-cells/4d4a769/ : FAILURE in 35m 25s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-full-ceph/deeedcc/ : SUCCESS in 39m 19s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/87/175787/2/check/check-tempest-dsvm-multinode-full/9b8b0c6/ : FAILURE in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ae11ae762419b81ce9f98f2e571d6b7e00fb308","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 16:00:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"0f6bba7353be2d46ee478235e7b01108c621072c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 18:45:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/87/175787/2/gate/gate-nova-docs/602d2aa//doc/build/html/ : SUCCESS in 11m 48s\n- gate-nova-pep8 http://logs.openstack.org/87/175787/2/gate/gate-nova-pep8/87d3255/ : SUCCESS in 8m 46s\n- gate-nova-python27 http://logs.openstack.org/87/175787/2/gate/gate-nova-python27/8cc4bd3/ : SUCCESS in 16m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-full/8f69498/ : SUCCESS in 40m 44s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-postgres-full/7713eeb/ : SUCCESS in 41m 26s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-neutron-full/796608b/ : SUCCESS in 54m 11s\n- gate-grenade-dsvm http://logs.openstack.org/87/175787/2/gate/gate-grenade-dsvm/47350a3/ : SUCCESS in 37m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-large-ops/e0244bd/ : SUCCESS in 23m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-neutron-large-ops/df00288/ : SUCCESS in 21m 51s\n- gate-nova-tox-functional http://logs.openstack.org/87/175787/2/gate/gate-nova-tox-functional/4f5e7a5/ : SUCCESS in 13m 22s\n- gate-devstack-dsvm-cells http://logs.openstack.org/87/175787/2/gate/gate-devstack-dsvm-cells/b42069c/ : SUCCESS in 20m 00s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/87/175787/2/gate/gate-grenade-dsvm-partial-ncpu/41de479/ : SUCCESS in 39m 11s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-nova-v21-full/6670f34/ : SUCCESS in 40m 44s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/87/175787/2/gate/gate-tempest-dsvm-full-ceph/5f47bb7/ : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9122028f4622b38fce6b58b625d81af52dd04519","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 18:45:54.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"880a356e40d327c0af4ce94b5a08fe0cd6fcab5d","revisions":{"bae71d5036e7cf239bfa9637c4f73c8773a8dcc1":{"kind":"REWORK","_number":1,"created":"2015-04-21 09:40:14.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/87/175787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/175787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/175787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/175787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/175787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/175787/1"}}},"commit":{"parents":[{"commit":"22d7547c6b62fb9dabd861e4941edd34eedabfc6","subject":"update .gitreview for stable/kilo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/22d7547c6b62fb9dabd861e4941edd34eedabfc6"}]}],"author":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2015-04-07 19:53:32.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@rackspace.com","date":"2015-04-21 09:39:48.000000000","tz":60},"subject":"scheduler: re-calculate NUMA on consume_from_instance","message":"scheduler: re-calculate NUMA on consume_from_instance\n\nThis patch narrows down the race window between the filter running and\nthe consumption of resources from the instance after the host has been\nchosen.\n\nIt does so by re-calculating the fitted NUMA topology just before consuming it\nfrom the chosen host. Thus we avoid any locking, but also make sure that\nthe host_state is kept as up to date as possible for concurrent\nrequests, as there is no opportunity for switching threads inside a\nconsume_from_instance.\n\nSeveral things worth noting:\n  * Scheduler being lock free (and thus racy) does not really affect\n  resources other than PCI and NUMA topology this badly - this is due\n  to complexity of said resources. In order for scheduler decesions to not\n  be based on basically guessing, in case of those two we will likely need\n  to introduce either locking or special heuristics.\n\n  * There is a lot of repeated code between the \u0027consume_from_instance\u0027\n  method and the actual filters. This situation should really be fixed but\n  is out of scope for this bug fix (which is about preventing valid\n  requests failing because of races in the scheduler).\n\nChange-Id: If0c7ad20506c9dddf4dec1eb64c9d6dd4fb75633\nCloses-bug: #1438238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bae71d5036e7cf239bfa9637c4f73c8773a8dcc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bae71d5036e7cf239bfa9637c4f73c8773a8dcc1"}]},"parents_data":[{"branch_name":"refs/heads/stable/kilo","commit_id":"22d7547c6b62fb9dabd861e4941edd34eedabfc6","is_merged_in_target_branch":true,"change_id":"I6356513ac42b79402dbde8ee5e75cbbd1aee7eef","change_number":174067,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/kilo"},"880a356e40d327c0af4ce94b5a08fe0cd6fcab5d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-04-21 09:50:59.000000000","uploader":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"ref":"refs/changes/87/175787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/175787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/175787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/175787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/175787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/175787/2"}}},"commit":{"parents":[{"commit":"22d7547c6b62fb9dabd861e4941edd34eedabfc6","subject":"update .gitreview for stable/kilo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/22d7547c6b62fb9dabd861e4941edd34eedabfc6"}]}],"author":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2015-04-07 19:53:32.000000000","tz":60},"committer":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2015-04-21 09:37:33.000000000","tz":60},"subject":"scheduler: re-calculate NUMA on consume_from_instance","message":"scheduler: re-calculate NUMA on consume_from_instance\n\nThis patch narrows down the race window between the filter running and\nthe consumption of resources from the instance after the host has been\nchosen.\n\nIt does so by re-calculating the fitted NUMA topology just before consuming it\nfrom the chosen host. Thus we avoid any locking, but also make sure that\nthe host_state is kept as up to date as possible for concurrent\nrequests, as there is no opportunity for switching threads inside a\nconsume_from_instance.\n\nSeveral things worth noting:\n  * Scheduler being lock free (and thus racy) does not really affect\n  resources other than PCI and NUMA topology this badly - this is due\n  to complexity of said resources. In order for scheduler decesions to not\n  be based on basically guessing, in case of those two we will likely need\n  to introduce either locking or special heuristics.\n\n  * There is a lot of repeated code between the \u0027consume_from_instance\u0027\n  method and the actual filters. This situation should really be fixed but\n  is out of scope for this bug fix (which is about preventing valid\n  requests failing because of races in the scheduler).\n\nChange-Id: If0c7ad20506c9dddf4dec1eb64c9d6dd4fb75633\nCloses-bug: #1438238\n(cherry picked from commit d6b3156a6c89ddff9b149452df34c4b32c50b6c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/880a356e40d327c0af4ce94b5a08fe0cd6fcab5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/880a356e40d327c0af4ce94b5a08fe0cd6fcab5d"}]},"parents_data":[{"branch_name":"refs/heads/stable/kilo","commit_id":"22d7547c6b62fb9dabd861e4941edd34eedabfc6","is_merged_in_target_branch":true,"change_id":"I6356513ac42b79402dbde8ee5e75cbbd1aee7eef","change_number":174067,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/kilo"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
