)]}'
{"id":"openstack%2Fnova~180151","triplet_id":"openstack%2Fnova~master~Ie6c768fc915553da73160ea51961078bfbacec77","project":"openstack/nova","branch":"master","topic":"bug/1451831","hashtags":[],"change_id":"Ie6c768fc915553da73160ea51961078bfbacec77","subject":"Correct memory validation for live migration","status":"MERGED","created":"2015-05-05 13:38:38.000000000","updated":"2016-02-04 14:01:18.000000000","submitted":"2015-11-16 13:51:18.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cbfd2d3f4be7dbaba8dfba4fdf606c13bd1839cf","_number":180151,"virtual_id_number":180151,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2015-10-06 17:41:26.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-10-06 17:24:31.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-11-16 13:51:18.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-10-06 23:27:15.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2015-10-05 13:43:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2015-10-05 14:11:31.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2016-02-04 14:01:18.000000000","post_submit":true,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"date":"2015-11-16 13:51:34.000000000","post_submit":true,"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},{"value":0,"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"date":"2015-10-05 23:03:13.000000000","_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},{"value":0,"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":1,"date":"2015-10-06 19:40:25.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"date":"2015-10-05 14:41:05.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":1,"date":"2015-10-07 15:56:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"date":"2015-10-05 16:21:42.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2015-10-06 17:28:15.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":1,"date":"2015-10-06 15:23:13.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},{"value":0,"date":"2015-10-05 18:46:38.000000000","_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},{"value":1,"date":"2015-10-06 14:03:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},{"value":2,"date":"2015-11-16 11:52:17.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2015-10-21 09:10:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},{"value":0,"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-11-16 02:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2015-11-15 22:51:56.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":1,"date":"2015-10-05 22:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2015-10-05 16:41:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},{"value":0,"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},{"value":1,"date":"2015-11-16 11:52:17.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},{"value":0,"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},{"value":0,"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},{"value":0,"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},{"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-07-02 19:51:17.000000000","updated_by":{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},"reviewer":{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},"state":"REVIEWER"},{"updated":"2015-09-17 15:33:34.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2015-09-17 15:49:20.000000000","updated_by":{"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},"reviewer":{"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},"state":"REVIEWER"},{"updated":"2015-09-17 20:29:05.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2015-09-22 02:27:00.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2015-10-05 13:43:57.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-05 14:11:31.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-05 14:41:05.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-10-05 16:21:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-05 16:41:59.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2015-10-05 22:41:13.000000000","updated_by":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"reviewer":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"state":"REVIEWER"},{"updated":"2015-10-05 23:03:13.000000000","updated_by":{"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},"reviewer":{"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},"state":"REVIEWER"},{"updated":"2015-10-06 14:03:37.000000000","updated_by":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"reviewer":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"state":"REVIEWER"},{"updated":"2015-10-06 15:23:13.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-06 17:24:31.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-06 17:28:15.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-06 17:41:26.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-06 19:40:25.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-06 23:27:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-07 15:56:54.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-21 09:10:22.000000000","updated_by":{"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},"reviewer":{"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},"state":"REVIEWER"},{"updated":"2015-11-15 22:51:56.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2015-11-16 02:17:48.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2015-11-16 11:52:17.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2015-11-16 13:51:18.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-11-16 13:51:34.000000000","updated_by":{"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},"reviewer":{"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},"state":"REVIEWER"},{"updated":"2016-02-04 14:01:18.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"}],"messages":[{"id":"0e84927a91c1b2181227ba82b03e0cc2b45db574","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-05-05 13:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c2ae76771e5f66cab4bcea9c81ff9eb4a4eb3a0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-05-05 13:40:21.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"9dc3eed64f8fc04667a9366a941684ce91dc4305","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-05-05 14:34:36.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/1/check-ibm-tempest-dsvm-full/1f36505 : SUCCESS in 51m 20s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/1/check-ibm-tempest-dsvm-postgres-full/e44043e : SUCCESS in 55m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c942ea01a7f269364a7e00350be775af501cb04","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-05 14:48:47.000000000","message":"Patch Set 1:\n\n* pci-test http://198.175.100.33/180151/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2a0f6bbbf9134c22a1c9ae499bddbe51a33100fa","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-05-05 16:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/180151/1/ : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e421ef3f415da3a97a15768547b88d7c3b5c05db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-05 21:34:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/1/check/gate-nova-pep8/d71b878/ : SUCCESS in 9m 57s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/1/check/gate-nova-docs/6087274//doc/build/html/ : SUCCESS in 11m 26s\n- gate-nova-python27 http://logs.openstack.org/51/180151/1/check/gate-nova-python27/97ad17c/ : FAILURE in 16m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-full/9a44bc7/ : FAILURE in 11m 57s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-postgres-full/4adfb1f/ : FAILURE in 15m 53s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-neutron-full/f6bebcd/ : FAILURE in 19m 16s\n- check-grenade-dsvm http://logs.openstack.org/51/180151/1/check/check-grenade-dsvm/feeacdd/ : FAILURE in 22m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/1/check/gate-tempest-dsvm-large-ops/5e75aaf/ : SUCCESS in 22m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/1/check/gate-tempest-dsvm-neutron-large-ops/4668e7e/ : SUCCESS in 17m 37s\n- check-devstack-dsvm-cells http://logs.openstack.org/51/180151/1/check/check-devstack-dsvm-cells/216a72c/ : FAILURE in 18m 18s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/1/check/gate-nova-tox-functional/fd25d1f/ : SUCCESS in 15m 01s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/1/check/check-grenade-dsvm-partial-ncpu/9a836bb/ : FAILURE in 22m 11s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-ironic-pxe_ssh/d4d4324/ : FAILURE in 13m 57s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-nova-v21-full/7b13ac3/ : FAILURE in 19m 21s\n- check-tempest-dsvm-cells http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-cells/fc40159/ : FAILURE in 14m 39s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-full-ceph/8ccd4b3/ : FAILURE in 37m 17s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-multinode-full/37724ff/ : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c323122dabc3e038851bbc3eba38c7da60682719","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-05-06 03:49:54.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/611a0ff6b4234cc7aceb0402835a7b3d/index.html : SUCCESS in 16m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/440f72fd5a954d068157cc0748e5b430/index.html : SUCCESS in 22m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/562fe98d8a634a84acd335347c2c87aa/index.html : SUCCESS in 19m 29s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_percona_user_001:th-percona/ff7d273c17004f74b782f65255c791b8/index.html : SUCCESS in 26m 24s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/75cc29659319480e846d27cbf16811e9/index.html : SUCCESS in 19m 51s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_percona_user_002:th-percona/f4b7c2e0eb0f450ab6de66175e5fdded/index.html : SUCCESS in 16m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"10f00e350353641ae4faa3b6bfff3b0e0af2a351","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-05-06 08:23:59.000000000","message":"Patch Set 1:\n\nrecheck bug 1451992","accounts_in_message":[],"_revision_number":1},{"id":"137a9dbde88c1928b2f4433f8c12c7b3d50f12ff","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-05-06 08:53:09.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9b2500c0b46b4d2586823caba575ddf7/index.html : SUCCESS in 16m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3ee703bb3b964e8a80ad126efc39875d/index.html : SUCCESS in 28m 46s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1e8578b2b6ed47aa8a332d2df46d846f/index.html : SUCCESS in 19m 56s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_percona_user_001:th-percona/e3bcccf8fd1c4b29a7f53c6c5cd01b96/index.html : SUCCESS in 26m 02s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/58fbe39d1ba2422dba1a423475e5bd23/index.html : SUCCESS in 17m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/1/check/real-db-upgrade_nova_percona_user_002:th-percona/dfdb2b6d4cc447a7b0a4fdb1e46bcfd6/index.html : SUCCESS in 16m 46s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"1cd65e4ff4e1e7bb40b9d68b2891af13741db22d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-06 09:35:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/1/check/gate-nova-pep8/ff062ba/ : SUCCESS in 15m 11s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/1/check/gate-nova-docs/287eb3d//doc/build/html/ : SUCCESS in 12m 29s\n- gate-nova-python27 http://logs.openstack.org/51/180151/1/check/gate-nova-python27/992ec68/ : FAILURE in 16m 29s\n- check-tempest-dsvm-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-full/394563a/ : SUCCESS in 43m 32s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-postgres-full/5a95b8f/ : SUCCESS in 52m 25s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-neutron-full/da31aaf/ : SUCCESS in 1h 06m 09s\n- check-grenade-dsvm http://logs.openstack.org/51/180151/1/check/check-grenade-dsvm/242494f/ : SUCCESS in 46m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/1/check/gate-tempest-dsvm-large-ops/c2131b2/ : SUCCESS in 30m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/1/check/gate-tempest-dsvm-neutron-large-ops/5df5150/ : SUCCESS in 29m 28s\n- check-devstack-dsvm-cells http://logs.openstack.org/51/180151/1/check/check-devstack-dsvm-cells/276201e/ : SUCCESS in 27m 25s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/1/check/gate-nova-tox-functional/73641d5/ : SUCCESS in 15m 28s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/1/check/check-grenade-dsvm-partial-ncpu/b9b0c5c/ : SUCCESS in 50m 00s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-ironic-pxe_ssh/13d36f0/ : SUCCESS in 40m 14s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-nova-v21-full/a087d91/ : SUCCESS in 44m 55s\n- check-tempest-dsvm-cells http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-cells/48b9db8/ : SUCCESS in 46m 28s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-full-ceph/46a921e/ : SUCCESS in 42m 17s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/1/check/check-tempest-dsvm-multinode-full/9290d30/ : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d440a94a53d7acfee67160209eb7d9b5c3a6083e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-05-06 09:57:27.000000000","message":"Patch Set 1: -Verified\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/180151/1/ : FAILURE in 1h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9bc2bd322f006113a98d884b93659b7b6bdc51d6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-05-06 11:16:41.000000000","message":"Patch Set 1:\n\nBuild failed. Logs: http://208.91.1.172/logs/180151/1 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/180151/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"039c43cb91694ac665e49daffb3604ef2604d609","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-05-06 13:45:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ff89aa46b89bd4f1349798508deca172fdd666b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-06 14:09:01.000000000","message":"Patch Set 2:\n\n* pci-test http://198.175.100.33/180151/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0d8c87865cf0083f644687a07ea19b78b6ef600b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-05-06 14:12:11.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3f06b9e61c1f4b1fa00cc08a2aa00327/index.html : SUCCESS in 17m 02s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2c7a8d1d63c7417ab78aa343a39e67e7/index.html : SUCCESS in 23m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/119c2f08bfd4446d8e1d921afd9ec350/index.html : SUCCESS in 19m 26s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/2/check/real-db-upgrade_nova_percona_user_001:th-percona/7edfe967d9c34255a0412eedeb9dc347/index.html : SUCCESS in 26m 20s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/681056f4cc94440a8a7d1d7233d28eeb/index.html : SUCCESS in 20m 08s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/2/check/real-db-upgrade_nova_percona_user_002:th-percona/9d7e401bf4ed4df3b39593bc8de46b2a/index.html : SUCCESS in 19m 58s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"910b0c234a9335e02cd0668258525e6f1222c713","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-05-06 14:38:03.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/2/check-ibm-tempest-dsvm-full/3bf157d : SUCCESS in 49m 10s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/2/check-ibm-tempest-dsvm-postgres-full/63862af : SUCCESS in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4689f1b40069c6bee2e79e3206e11a100519ce7f","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-05-06 15:12:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a0153c6b8cbea88f7da3bdc65b7ef2b04d696db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-06 15:14:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/2/check/gate-nova-pep8/aa38755/ : SUCCESS in 11m 46s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/2/check/gate-nova-docs/9b924e0//doc/build/html/ : SUCCESS in 12m 00s\n- gate-nova-python27 http://logs.openstack.org/51/180151/2/check/gate-nova-python27/cd14c7d/ : SUCCESS in 11m 53s\n- check-tempest-dsvm-full http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-full/f2b7b5c/ : SUCCESS in 46m 57s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-postgres-full/e607729/ : SUCCESS in 49m 29s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-neutron-full/09ba25f/ : SUCCESS in 1h 03m 47s\n- check-grenade-dsvm http://logs.openstack.org/51/180151/2/check/check-grenade-dsvm/bb5bf5b/ : SUCCESS in 39m 15s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/2/check/gate-tempest-dsvm-large-ops/6631446/ : SUCCESS in 23m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/2/check/gate-tempest-dsvm-neutron-large-ops/41ed164/ : SUCCESS in 22m 01s\n- check-devstack-dsvm-cells http://logs.openstack.org/51/180151/2/check/check-devstack-dsvm-cells/006b7c4/ : SUCCESS in 19m 24s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/2/check/gate-nova-tox-functional/d8a66c9/ : SUCCESS in 11m 33s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/2/check/check-grenade-dsvm-partial-ncpu/e0113f1/ : SUCCESS in 42m 30s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-ironic-pxe_ssh/7b90394/ : SUCCESS in 31m 02s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-nova-v21-full/b2a70d3/ : SUCCESS in 37m 04s\n- check-tempest-dsvm-cells http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-cells/f230b2e/ : SUCCESS in 37m 43s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-full-ceph/8f5570d/ : SUCCESS in 38m 32s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/2/check/check-tempest-dsvm-multinode-full/81cf618/ : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65b4307da10afc6e92a915ae8614b9b6a07c476a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-05-06 17:11:02.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/180151/2/ : FAILURE in 1h 08m 11s","accounts_in_message":[],"_revision_number":2},{"id":"67f8795695c2a742cca58e191713a64b936ed0ea","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-05-07 10:15:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/2/commentonly/dsvm-tempest-xen/e0abf58 : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6c1f7be8fa958f2912c980090312c40b8db95953","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-05-07 23:06:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/180151/2 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/180151/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bc80bd8011319da5af5e06ad189a429cf57b3237","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-11 16:53:33.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"df386548407de81c9b0b9edf16361c6e88174f7b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-05-26 13:08:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b1c64cd578c1cc3bc09049616f2909d733da0ae","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-05-26 13:09:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"8995d2675f3ce3be82eb683b629a5dc3589b348b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-05-26 13:36:31.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/206e8d76203a475884364bf9468d28f3/index.html : SUCCESS in 17m 30s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ae83b378d57f4ce1989d10e81ef87b9e/index.html : SUCCESS in 27m 13s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5951fe248b8b4702b302a82118750c13/index.html : SUCCESS in 26m 38s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/3/check/real-db-upgrade_nova_percona_user_001:th-percona/56e794eb9f4546f39a5f9d77d2795d69/index.html : SUCCESS in 27m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1acd9922531043f6a5a9d211897b6f48/index.html : SUCCESS in 18m 37s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/3/check/real-db-upgrade_nova_percona_user_002:th-percona/1fc0ff53738d43b7984a38568504f65a/index.html : SUCCESS in 20m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"8f282a6f730079d5eeb4a473120f0cf033602707","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-05-26 13:40:00.000000000","message":"Patch Set 3:\n\n* pci-test http://198.175.100.33/180151/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"eb80f2a5bbee5ff7cb19fe3319b2f2cf53df6a8e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-05-26 13:56:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/3/check/gate-nova-pep8/85eeb16/ : SUCCESS in 8m 35s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/3/check/gate-nova-docs/ce3f422//doc/build/html/ : SUCCESS in 7m 47s\n- gate-nova-python27 http://logs.openstack.org/51/180151/3/check/gate-nova-python27/4b0a109/ : SUCCESS in 12m 28s\n- check-tempest-dsvm-full http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-full/f0e46f4/ : SUCCESS in 37m 31s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-postgres-full/cd5a9b2/ : SUCCESS in 35m 04s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-neutron-full/12ccd05/ : SUCCESS in 43m 32s\n- check-grenade-dsvm http://logs.openstack.org/51/180151/3/check/check-grenade-dsvm/72e8062/ : SUCCESS in 39m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/3/check/gate-tempest-dsvm-large-ops/8959c92/ : SUCCESS in 27m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/3/check/gate-tempest-dsvm-neutron-large-ops/267a7e5/ : SUCCESS in 23m 44s\n- check-devstack-dsvm-cells http://logs.openstack.org/51/180151/3/check/check-devstack-dsvm-cells/b3165e0/ : SUCCESS in 16m 24s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/3/check/gate-nova-tox-functional/1a5d922/ : SUCCESS in 11m 13s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/3/check/check-grenade-dsvm-partial-ncpu/8da60fa/ : SUCCESS in 38m 00s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-ironic-pxe_ssh/875b769/ : SUCCESS in 27m 32s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-nova-v21-full/5e614a9/ : SUCCESS in 44m 45s\n- check-tempest-dsvm-cells http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-cells/107a65b/ : SUCCESS in 37m 30s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-full-ceph/71c3e2d/ : SUCCESS in 45m 28s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/3/check/check-tempest-dsvm-multinode-full/2617131/ : SUCCESS in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c14386456e5174d9aa30da74610d2ad8ea753ea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-05-26 14:13:14.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/3/check-ibm-tempest-dsvm-full/23eb8b7 : SUCCESS in 58m 06s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/3/check-ibm-tempest-dsvm-postgres-full/5771d7f : SUCCESS in 1h 04m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72ba0d7b7110d853262d2e7a708653258096a639","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-05-26 14:37:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/3/check/dsvm-tempest-xen/7f536a4 : SUCCESS in 1h 28m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c723e033f2d23a8173b8835b7f2f717a7bbdff02","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-05-26 14:54:13.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/180151/3/ : FAILURE in 1h 20m 02s","accounts_in_message":[],"_revision_number":3},{"id":"5fd157285f1457fd50d9da81e951cc2947bc9c6e","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-05-26 20:42:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bfb28faf0820816f94fa38fb33957b705d44e689","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-05-27 22:12:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/180151/3 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/180151/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5baa30639414d1afbdfae630188bc2624c21cd0b","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-06-18 10:24:28.000000000","message":"Patch Set 3: Code-Review-1\n\nAs we\u0027ve discussed in IRC, this patch is workaround, until cool BPs that recover single responsibility principle of scheduling come on the stage. So while this principle was corrupted by design, correct redesign, as for me, would be to duplicate scheduler logic, related with alloc ratios, in conductor instead.","accounts_in_message":[],"_revision_number":3},{"id":"34a33e982fda71c644a345e620b5072c336567c2","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-06-22 12:55:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\ncan\u0027t we just move the check into the compute node call that we already make:\nhttps://github.com/openstack/nova/blob/master/nova/compute/manager.py#L4740","accounts_in_message":[],"_revision_number":3},{"id":"649841e5458bf3fe57657f35e83f3dfc56433cb0","author":{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},"date":"2015-07-02 19:51:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2655aec13088ae3ef60780cd342abeb2e4cbb17d","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-07-02 21:17:04.000000000","message":"Patch Set 3:\n\nJohn Garbutt, I don\u0027t think these check are the same.","accounts_in_message":[],"_revision_number":3},{"id":"ed9f7abea57a6bbec3ca7ecd545a0435801daeff","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-17 15:23:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"08d813dcdc5a80483f0d0814516316ee42b458b3","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-09-17 15:32:01.000000000","message":"Patch Set 4: Code-Review+1\n\nthis looks better:) \nnit: Closes bug in commit-message is referencing to duplicate, is it OK?","accounts_in_message":[],"_revision_number":4},{"id":"4b48f04cbe18959fc4dafcb1abbc9b95f6cdc8bd","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-09-17 15:33:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dca7f963d7e803c23e1acf17e6e01df5021544a0","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-17 15:43:57.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3c2f3e181a3a4e77bb24d3b893d222bb/index.html : SUCCESS in 12m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/70679dda5d9840dfa87fb8d790bd5a4f/index.html : SUCCESS in 14m 22s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1199863283df46ee8e4afff0fa8879fe/index.html : SUCCESS in 16m 03s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/4/check/real-db-upgrade_nova_percona_user_001:th-percona/29457c7d39684e80a2142499252f78ee/index.html : SUCCESS in 15m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4d8df1f5d6c04be99cd9aeb8aab7714c/index.html : SUCCESS in 13m 59s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/4/check/real-db-upgrade_nova_percona_user_002:th-percona/92cad52923fe4856bfe692b6a74d2ad6/index.html : SUCCESS in 14m 29s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"4ba8e8b26ba0955517c028180fa1116d164b8070","author":{"_account_id":6542,"name":"j_king","email":"james@agentultra.com","username":"james-agentultra"},"date":"2015-09-17 15:49:20.000000000","message":"Patch Set 4: Code-Review-1\n\nI agree w/ garyk\u0027s comment, +1 otherwise. Looks good. Thanks for the patch. :)","accounts_in_message":[],"_revision_number":4},{"id":"6d8d6bcce80244f8bd877c2cdeb9f7ff66e39fff","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-09-17 15:52:15.000000000","message":"Patch Set 4:\n\nSylvain, maybe it would be better to reuse scheduler then first, as we\u0027ve discussed morning? It wouldn\u0027t be great until we have persistent request_spec, but it seems even better.","accounts_in_message":[],"_revision_number":4},{"id":"92cadce84eafecb9ff4baa898a0b19422139e160","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-17 15:52:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae192fd2fd95a791491b7474d911d7c5eef53659","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-17 15:56:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2540e797d3020252329f645c892e46be102e7d67","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-17 15:56:44.000000000","message":"Patch Set 5:\n\n@tdurakov, the strategy is explained in http://specs.openstack.org/openstack/nova-specs/specs/liberty/approved/check-destination-on-migrations.html where we had consensus on that","accounts_in_message":[],"_revision_number":5},{"id":"a3fda8b914beaa9dfc77b8113d63642132207a8d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-17 16:13:16.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/804c31d16c2a4db7935d1735ec98c885/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2e29b48bb866444e8529ee570a53649c/index.html : SUCCESS in 13m 32s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5d12a0b1c0644445b2f2b78abcf887f6/index.html : SUCCESS in 14m 15s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/5/check/real-db-upgrade_nova_percona_user_001:th-percona/f2bf86a22af24c4b9e914bc928f541d2/index.html : SUCCESS in 16m 27s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ac332bdbb5c144e2959239efe7ce0271/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/5/check/real-db-upgrade_nova_percona_user_002:th-percona/cfe92265edf4462fbef0327f7fba4cb9/index.html : SUCCESS in 14m 56s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"bcc519ad765e31a41a18dafa3597e523d73458b8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-17 16:18:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3e55a117bf996c7f1eb8198f08b338a5f7602f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-17 16:51:20.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/5/check-ibm-tempest-dsvm-full/ddc1152 : FAILURE in 58m 16s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/5/check-ibm-tempest-dsvm-postgres-full/9a1b7a9 : FAILURE in 53m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"36ca62347651cce5f2ec2b0e83165cfecd3df926","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-17 16:59:32.000000000","message":"Patch Set 5: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/51/180151/5/11924/index.html : FAILURE in 01 h 04 m 33 s","accounts_in_message":[],"_revision_number":5},{"id":"c3d3041a8f39da2d2e9fbaab24aad39eb57afd7d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-17 17:32:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/180151/5/ : SUCCESS in 1h 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d114d2529039e74c4ea8fa17c4dc6ae4818b1855","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-17 18:14:46.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/180151/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"fb57527beec885700c89c1b61b78da3ef6fb05cc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-17 18:23:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/5/check/dsvm-tempest-xen/7b722b9 : FAILURE in 1h 16m 13s","accounts_in_message":[],"_revision_number":5},{"id":"c3cec1b72fe94693d190c1ab29524b9184871300","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-17 20:04:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/180151/5/Nova-ML2-Sriov : SUCCESS in 43m 51s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"ac457bbfbbaa54b0a71cb92a57495226d68f8b89","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-09-17 20:29:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0139d4ecd0cd61ccf92b1c4cd1fa7feddbf06d3d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-17 22:12:54.000000000","message":"Patch Set 5:\n\nBuild failed. Logs: http://208.91.1.172/logs/180151/5 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/180151/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"3f94516bd79a8425cb414b5a3a0c0030a08c63d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-17 23:57:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/5/check/gate-nova-pep8/06b861e/ : SUCCESS in 9m 33s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/5/check/gate-nova-docs/93836b8//doc/build/html/ : SUCCESS in 14m 17s\n- gate-nova-python27 http://logs.openstack.org/51/180151/5/check/gate-nova-python27/d03a2ab/ : SUCCESS in 15m 06s\n- gate-nova-python34 http://logs.openstack.org/51/180151/5/check/gate-nova-python34/5ad2dbc/ : SUCCESS in 13m 20s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-full/f1a6631/ : SUCCESS in 59m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-postgres-full/fbac97b/ : SUCCESS in 40m 54s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-neutron-full/d93be09/ : SUCCESS in 1h 16m 39s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/5/check/gate-grenade-dsvm/fed52e3/ : SUCCESS in 52m 11s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-large-ops/26c039a/ : SUCCESS in 33m 25s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-neutron-large-ops/91fef5f/ : SUCCESS in 33m 22s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/5/check/gate-nova-tox-functional/1ff512e/ : SUCCESS in 15m 37s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/5/check/gate-grenade-dsvm-partial-ncpu/92b6952/ : SUCCESS in 46m 57s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/afb9d84/ : SUCCESS in 52m 05s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-nova-v20-api/4f39824/ : SUCCESS in 41m 55s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-nova-v20-api-legacy/7a9db5b/ : SUCCESS in 41m 31s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-cells/7651e94/ : SUCCESS in 50m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-full-ceph/92328c1/ : SUCCESS in 43m 56s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/5/check/gate-tempest-dsvm-multinode-full/7ec3af3/ : FAILURE in 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e091d55c43d6008282551734bc6632e12f160031","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-18 00:20:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/51/180151/5/check/check-dsvm-vz7-tempest-exe-minimal/00d4f96/ : SUCCESS in 22m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"ad094d29ab126863064f0239aaa2b6d9fea43323","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-18 00:50:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/51/180151/5/check/dsvm-tempest-cinder/a77fda7/ : SUCCESS in 52m 24s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"824b6ea6fa78615ea73db24eb69608a5f972f90c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-18 15:11:48.000000000","message":"Patch Set 5:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":5},{"id":"d844c1291e20b7c1833b91afe34fa0ea5c01f2cf","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-09-21 14:18:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2a7607fe2f1c1987485d7c3e577d175300b5a059","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-21 14:43:45.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8178c153812d4d62b734bd617e599635/index.html : SUCCESS in 12m 20s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0394656bebb14525be364d3b533825a5/index.html : SUCCESS in 15m 22s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e8fa0760bd3940cbaf308fe5d309c5f4/index.html : SUCCESS in 13m 56s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/6/check/real-db-upgrade_nova_percona_user_001:th-percona/31d744db2b2f4d2eaa6db828b2b7fd7f/index.html : SUCCESS in 15m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c6dcde20c3440719fcb8e4058a7aaa3/index.html : SUCCESS in 11m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/6/check/real-db-upgrade_nova_percona_user_002:th-percona/de50b1a787224796b3cae258107c0b78/index.html : SUCCESS in 14m 27s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"32be8619c28669919eff7ab4db21acef22b520c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-21 15:18:53.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/6/check-ibm-tempest-dsvm-full/36f5ee9 : SUCCESS in 59m 34s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/180151/6/check-ibm-tempest-dsvm-postgres-full/d31e5ba : SUCCESS in 55m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b0e39afa116ee7edde80b9786b20408f63e0564d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-21 15:35:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/6/check/gate-nova-pep8/d86d2d8/ : SUCCESS in 9m 14s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/6/check/gate-nova-docs/ed85f8a//doc/build/html/ : SUCCESS in 9m 45s\n- gate-nova-python27 http://logs.openstack.org/51/180151/6/check/gate-nova-python27/a40ca77/ : SUCCESS in 10m 44s\n- gate-nova-python34 http://logs.openstack.org/51/180151/6/check/gate-nova-python34/0b2ed55/ : SUCCESS in 13m 26s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-full/2eeacc4/ : SUCCESS in 43m 39s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-postgres-full/808d434/ : SUCCESS in 50m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-neutron-full/5b8f888/ : SUCCESS in 59m 17s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/6/check/gate-grenade-dsvm/ee368d7/ : SUCCESS in 40m 45s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-large-ops/0facc40/ : SUCCESS in 33m 45s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-neutron-large-ops/d436fc2/ : SUCCESS in 35m 01s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/6/check/gate-nova-tox-functional/a4b0987/ : SUCCESS in 16m 29s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/6/check/gate-grenade-dsvm-partial-ncpu/e43fd50/ : SUCCESS in 47m 16s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ebd3637/ : FAILURE in 40m 29s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-nova-v20-api/f2f5927/ : SUCCESS in 42m 35s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-nova-v20-api-legacy/bbe1cd4/ : SUCCESS in 43m 34s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-cells/1f0615e/ : SUCCESS in 39m 28s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-full-ceph/11017b9/ : SUCCESS in 47m 23s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/6/check/gate-tempest-dsvm-multinode-full/2a908ed/ : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e93049ed21a0c90f999089f8601f0a72f1514c22","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2015-09-21 15:47:50.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nGood change; just needs another test.","accounts_in_message":[],"_revision_number":6},{"id":"fe829a4e1a6f9c4f0bebbc79070db3df0ae74c8d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-21 15:59:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/51/180151/6/check/check-dsvm-vz7-tempest-exe-minimal/db14873/ : SUCCESS in 22m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"8662e2d5b7b7fc111abe03902560f642a3b9dd10","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-21 16:07:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/180151/6/ : SUCCESS in 1h 47m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bbcbd3724e603b53d58e31d26afc77bd501addf0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-21 16:12:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/180151/6 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/180151/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"8968832ff54d9fd3d94f71ac967cba6271784d78","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-21 16:51:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/51/180151/6/check/dsvm-tempest-cinder/31011c1/ : FAILURE in 1h 15m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"06191cd9de88792856188668e1a68648956c79a5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-21 17:13:17.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/180151/6/Nova-ML2-Sriov : SUCCESS in 48m 44s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"a177beaaa78e04182805f23f617ae6a57c4d6ff4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-10-05 13:17:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"789bf716946da3d141844d016355f0fb51dc9391","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-10-05 13:35:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"46987af02e60834b1938ff6e9435e58288a3fd49","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-05 13:43:57.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/51/180151/7/check/check-ibm-tempest-dsvm-full/482f3e2/ : FAILURE in 7m 53s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/51/180151/7/check/check-ibm-tempest-dsvm-postgres-full/3ba2be1/ : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7b3d673f5a09a3fbb324f0f0e36b7db1ae008b53","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-05 14:06:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/7/check/dsvm-tempest-xen/10242f8 : FAILURE in 30m 30s","accounts_in_message":[],"_revision_number":7},{"id":"a137474ab6c8b35cc51c25a170bf6bad518d7bd5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-05 14:11:31.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/180151/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"c486a00d48d81ef2ce0cc140af1005d475c04918","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-05 14:39:48.000000000","message":"Patch Set 7: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/51/180151/7/12919/index.html : FAILURE in 01 h 01 m 24 s","accounts_in_message":[],"_revision_number":7},{"id":"534e5ba423983836ec6d024db928dfe59a4f873f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-05 14:40:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/7/check/gate-nova-pep8/577587f/ : SUCCESS in 8m 09s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/7/check/gate-nova-docs/35cf19b//doc/build/html/ : SUCCESS in 15m 02s\n- gate-nova-python27 http://logs.openstack.org/51/180151/7/check/gate-nova-python27/6b1ad73/ : SUCCESS in 12m 31s\n- gate-nova-python34 http://logs.openstack.org/51/180151/7/check/gate-nova-python34/c1a1403/ : SUCCESS in 10m 55s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full/bacdf67/ : SUCCESS in 1h 02m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-postgres-full/d19183c/ : SUCCESS in 45m 47s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-full/a45033f/ : SUCCESS in 1h 02m 12s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm/275997c/ : FAILURE in 48m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-large-ops/e4e183f/ : SUCCESS in 42m 15s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-large-ops/fea7274/ : SUCCESS in 36m 46s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/7/check/gate-nova-tox-functional/0e9f07e/ : SUCCESS in 16m 11s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm-partial-ncpu/cef71b2/ : SUCCESS in 57m 12s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b1dbc57/ : SUCCESS in 46m 57s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api/b8d2922/ : SUCCESS in 46m 10s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api-legacy/1a533d3/ : SUCCESS in 46m 30s\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-cells/87d858b/ : SUCCESS in 50m 49s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full-ceph/4a7675a/ : SUCCESS in 46m 10s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-multinode-full/3124c62/ : FAILURE in 58m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f4a3e9f31ccf0e5d3fe25cc9cce4cd62a764933a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-05 14:41:05.000000000","message":"Patch Set 7:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm: https://bugs.launchpad.net/bugs/1446583\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":7},{"id":"0550540c822005f9f582483220cedde2086cbb34","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-05 15:10:52.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/180151/7/ : SUCCESS in 1h 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a5da5cd4878453ada376b7c8286362098dda50de","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-05 16:21:42.000000000","message":"Patch Set 7:\n\nBuild failed. Logs: http://208.91.1.172/logs/180151/7 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/180151/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"6abfadea5196b66064cfd184c7a38c986564151f","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2015-10-05 16:41:59.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e1bc88ea1200e93f39628f6b74b18de266c2500e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-05 17:46:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/180151/7/Nova-ML2-Sriov : SUCCESS in 53m 17s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"167cb2a0229fbbb79fb8983367ccea2b4aac858a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-10-05 18:46:38.000000000","message":"Patch Set 7:\n\nrecheck bug 1446583","accounts_in_message":[],"_revision_number":7},{"id":"7bd6e0075ef7c08b6fe3234041349a6f30fa9229","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-05 20:01:54.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/7/check/gate-nova-pep8/524adef/ : SUCCESS in 7m 16s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/7/check/gate-nova-docs/a26fa0e//doc/build/html/ : SUCCESS in 14m 35s\n- gate-nova-python27 http://logs.openstack.org/51/180151/7/check/gate-nova-python27/19c4b5e/ : SUCCESS in 13m 48s\n- gate-nova-python34 http://logs.openstack.org/51/180151/7/check/gate-nova-python34/7379669/ : SUCCESS in 8m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full/ece571a/ : SUCCESS in 53m 56s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-postgres-full/e72794b/ : SUCCESS in 52m 17s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-full/6d9a79e/ : SUCCESS in 1h 05m 46s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm/8115755/ : FAILURE in 53m 58s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-large-ops/ebd39bb/ : SUCCESS in 37m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-large-ops/479cb45/ : SUCCESS in 36m 21s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/7/check/gate-nova-tox-functional/1e9c124/ : SUCCESS in 17m 34s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm-partial-ncpu/d63519c/ : SUCCESS in 50m 56s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/017f4d3/ : SUCCESS in 35m 12s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api/1449862/ : SUCCESS in 48m 46s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api-legacy/4b6e3f0/ : SUCCESS in 40m 17s\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-cells/c164c57/ : SUCCESS in 51m 15s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full-ceph/79f654b/ : SUCCESS in 55m 49s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-multinode-full/d47ce38/ : SUCCESS in 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0da53b52a606e2e04508d7de9796cd8fc90fe027","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-05 20:02:10.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/7/check/dsvm-tempest-xen/ff567e6 : FAILURE in 1h 15m 07s","accounts_in_message":[],"_revision_number":7},{"id":"08eeb5e3fd77760e2c540429bb087b252c7ca937","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-05 21:17:39.000000000","message":"Patch Set 7:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm: https://bugs.launchpad.net/bugs/1446583\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":7},{"id":"aabce87ce5a70959b19680e3a4fb764700c86e99","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2015-10-05 22:41:13.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"85356ecfc1dabf2603948726b5c4630f0bb7751c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-05 23:01:36.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/180151/7/ : FAILURE in 3h 00m 02s","accounts_in_message":[],"_revision_number":7},{"id":"156ebef9c6525c768bc418a6143f453e43128a3b","author":{"_account_id":10635,"name":"Octavian Ciuhandu","email":"ociuhandu@cloudbasesolutions.com","username":"ociuhandu"},"date":"2015-10-05 23:03:13.000000000","message":"Patch Set 7:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":7},{"id":"3bcbb7810a82700de75e855dfd12a29b78ceec40","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-06 00:23:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/180151/7/Nova-ML2-Sriov : SUCCESS in 53m 22s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"ad8eca299ecc0fa16d628fa23a5ed5fefa74ac8f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-06 05:23:20.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/180151/7/ : FAILURE in 2h 03m 19s","accounts_in_message":[],"_revision_number":7},{"id":"c3af1f3f5608b091f7a373fdc6f4f6406f48d9b7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-10-06 13:59:16.000000000","message":"Patch Set 7:\n\nrecheck bug 1446583","accounts_in_message":[],"_revision_number":7},{"id":"8b16560b434ce55b0c7f3493bb53ddeab45606ca","author":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"date":"2015-10-06 14:03:37.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"11a5fa33f1dab2d438f333fffc8038a108cfbc4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-06 15:19:30.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/7/check/gate-nova-pep8/c1ab81a/ : SUCCESS in 9m 55s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/7/check/gate-nova-docs/bf4a593//doc/build/html/ : SUCCESS in 14m 09s\n- gate-nova-python27 http://logs.openstack.org/51/180151/7/check/gate-nova-python27/4bd4b4b/ : SUCCESS in 14m 34s\n- gate-nova-python34 http://logs.openstack.org/51/180151/7/check/gate-nova-python34/4fa6351/ : SUCCESS in 13m 24s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full/2139e3e/ : SUCCESS in 55m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-postgres-full/d029dfe/ : SUCCESS in 57m 20s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-full/9f81da0/ : SUCCESS in 1h 16m 24s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm/5ac0fe8/ : SUCCESS in 50m 04s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-large-ops/b0d8024/ : SUCCESS in 35m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-large-ops/4c00cf7/ : SUCCESS in 36m 29s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/7/check/gate-nova-tox-functional/68df53a/ : SUCCESS in 19m 31s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm-partial-ncpu/436ea73/ : SUCCESS in 1h 00m 42s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f02d348/ : SUCCESS in 37m 04s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api/f318c35/ : FAILURE in 47m 59s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api-legacy/185f204/ : SUCCESS in 46m 44s\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-cells/55ccf53/ : SUCCESS in 54m 03s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full-ceph/58776d7/ : SUCCESS in 1h 00m 09s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-multinode-full/f5e2cff/ : SUCCESS in 54m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fcafb219f00a48745f8abc7f20cbaf25a2529da6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-06 15:23:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/7/check/dsvm-tempest-xen/888e667 : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":7},{"id":"3e1ea999a7ed04b98769a23bce0f208b11459337","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-06 15:25:31.000000000","message":"Patch Set 7:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-nova-v20-api: https://bugs.launchpad.net/bugs/1501366\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":7},{"id":"943dd69a9d715f8000ad943f2bf5c6952e7b15cd","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-10-06 15:40:43.000000000","message":"Patch Set 7:\n\nrecheck bug 1501366","accounts_in_message":[],"_revision_number":7},{"id":"19b68dc876fcec8518ce9ba3a9bce86b39fa33d8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-06 16:49:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/51/180151/7/check/gate-nova-pep8/6c8b44c/ : SUCCESS in 7m 37s\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/7/check/gate-nova-docs/f5d13ff//doc/build/html/ : SUCCESS in 9m 36s\n- gate-nova-python27 http://logs.openstack.org/51/180151/7/check/gate-nova-python27/7458996/ : SUCCESS in 15m 07s\n- gate-nova-python34 http://logs.openstack.org/51/180151/7/check/gate-nova-python34/ae20aad/ : SUCCESS in 9m 28s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full/8919335/ : SUCCESS in 42m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-postgres-full/019fe79/ : SUCCESS in 43m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-full/f1f1f07/ : SUCCESS in 56m 26s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm/4d6757a/ : SUCCESS in 46m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-large-ops/e073f64/ : SUCCESS in 29m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-neutron-large-ops/90b9b7c/ : SUCCESS in 32m 13s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/7/check/gate-nova-tox-functional/d322e72/ : SUCCESS in 17m 40s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/51/180151/7/check/gate-grenade-dsvm-partial-ncpu/0444bbc/ : SUCCESS in 48m 21s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0727996/ : SUCCESS in 40m 56s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api/7f75319/ : SUCCESS in 37m 36s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-nova-v20-api-legacy/f8909c6/ : SUCCESS in 39m 24s\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-cells/b81bec6/ : SUCCESS in 38m 34s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-full-ceph/ee6a486/ : SUCCESS in 45m 33s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/51/180151/7/check/gate-tempest-dsvm-multinode-full/1ac729d/ : FAILURE in 59m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ad775351939f08550b2f8a9e3ffa92181e125b14","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-06 16:57:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/51/180151/7/check/dsvm-tempest-xen/b69d19e : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":7},{"id":"baa868e5301efc2a796cf4d03609ee5582203105","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-10-06 17:24:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/51/180151/7/check/check-dsvm-vz7-tempest-exe-minimal/44f0e8b/ : FAILURE in 33m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"d3b65ec5eda9c81b9fd8f610802e1b5f5dee5d87","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-06 17:28:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/180151/7/ : SUCCESS in 1h 41m 50s","accounts_in_message":[],"_revision_number":7},{"id":"7a25a2c80fbb85465c2372a6842e76b1b6183a72","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-10-06 17:41:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/51/180151/7/check/dsvm-tempest-cinder/cb97d6c/ : SUCCESS in 50m 05s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"faf72fc91d55c60e7a3537199b6f9f725ff26a99","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2015-10-06 18:34:44.000000000","message":"Patch Set 7:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":7},{"id":"f8b75429dcdc269faea7468c52ef6ae42fed912c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-06 19:40:25.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/51/180151/7/13045/index.html : SUCCESS in 01 h 01 m 46 s","accounts_in_message":[],"_revision_number":7},{"id":"159697c7a5768375d654cd55eb1f4662def13400","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-06 23:27:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/180151/7/Nova-ML2-Sriov : SUCCESS in 48m 30s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"93c56f3d55930659b5b652ad1274f4c31952d843","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-07 15:56:54.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/28ee1a211c9648d4a345b8c2c75f1c19/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c78a7f284a7c4cdfbb8f22b0aebbc529/index.html : SUCCESS in 14m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5a8d3c594fac4bc4b67267a4f3f64589/index.html : SUCCESS in 15m 23s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/7/check/real-db-upgrade_nova_percona_user_001:th-percona/bc12f567045a425fa72e817f326057f4/index.html : SUCCESS in 16m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/51/180151/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5f29d68b71af4e0fb697312f8f44e22d/index.html : SUCCESS in 14m 34s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/51/180151/7/check/real-db-upgrade_nova_percona_user_002:th-percona/6e4a789f7aed45f78b20c5fcf0a00013/index.html : SUCCESS in 15m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"034dc8823c55fbdee56e03f863d6252eafcf7547","author":{"_account_id":12814,"name":"Bartosz Fic","email":"bartosz.fic@intel.com","username":"bartosz-fic"},"date":"2015-10-21 09:10:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"9ffb05e430f9c280c537cbd26c5e5afe8a2b40b3","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2015-11-15 22:51:56.000000000","message":"Patch Set 7: Code-Review+2\n\nYay!","accounts_in_message":[],"_revision_number":7},{"id":"8e9bfb45bc369741bc975b747ff60d0bad9da050","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2015-11-16 02:17:48.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"1dcaa960a41707cce39b3687476c029cb63e0346","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-11-16 11:52:17.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"760e0bc828e55b2de2d68d4490f5ed4e9113a310","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-16 12:01:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"040ab102640f5e54692191b4c82582578d5f36c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-16 13:51:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/51/180151/7/gate/gate-nova-docs/acadfa9//doc/build/html/ : SUCCESS in 9m 44s\n- gate-nova-pep8 http://logs.openstack.org/51/180151/7/gate/gate-nova-pep8/55897b9/ : SUCCESS in 8m 38s\n- gate-nova-python27 http://logs.openstack.org/51/180151/7/gate/gate-nova-python27/c5372d9/ : SUCCESS in 14m 34s\n- gate-nova-python34 http://logs.openstack.org/51/180151/7/gate/gate-nova-python34/593180a/ : SUCCESS in 13m 18s\n- gate-tempest-dsvm-full http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-full/1dbc95a/ : SUCCESS in 53m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-postgres-full/4d1aaf2/ : SUCCESS in 55m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-neutron-full/7eddb0d/ : SUCCESS in 1h 14m 09s\n- gate-grenade-dsvm http://logs.openstack.org/51/180151/7/gate/gate-grenade-dsvm/2b584ba/ : SUCCESS in 59m 28s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-large-ops/8f06e63/ : SUCCESS in 49m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-neutron-large-ops/45e41d4/ : SUCCESS in 38m 00s\n- gate-nova-tox-functional http://logs.openstack.org/51/180151/7/gate/gate-nova-tox-functional/2727c49/ : SUCCESS in 21m 11s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/51/180151/7/gate/gate-grenade-dsvm-multinode/ef0ba1b/ : SUCCESS in 1h 07m 36s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-full-ceph/2e2f2af/ : SUCCESS in 55m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/51/180151/7/gate/gate-tempest-dsvm-cells/5228e2a/ : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c91e028da20c741d1a51185c890805cee8a9442b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-16 13:51:23.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":7},{"id":"3651519336773037544c43bf6e02cbab8cfa2a81","author":{"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},"date":"2015-11-16 13:51:34.000000000","message":"Patch Set 7:\n\nBuild Started http://openstratus-gerrit-3993.phx-os1.stratus.dev.ebay.com:9090/job/test-gerrit-plugin-2/3/","accounts_in_message":[],"_revision_number":7},{"id":"3286d660b9b90764f010e8ef8990e91356bc1094","author":{"_account_id":11466,"name":"ting zhang","email":"tingzhang@ebay.com","username":"tingzhang"},"date":"2015-11-16 13:51:35.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nhttp://openstratus-gerrit-3993.phx-os1.stratus.dev.ebay.com:9090/job/test-gerrit-plugin-2/3/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"71f2558bd7a2071666e9ec6f2d87c8aa73829557","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2016-02-04 14:01:18.000000000","message":"Patch Set 7:\n\nThis change now does correct math for memory validation but without claims - it\u0027s still racy. What we really want is to add claims to live migration and also make sure we are always sending the limits to the compute node.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f5e35dcfe8ee586106438bcfa551426babc75bf6","revisions":{"2f080d42b9617e22f17146b27655111c8759c9d4":{"kind":"REWORK","_number":1,"created":"2015-05-05 13:38:38.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/1"}}},"commit":{"parents":[{"commit":"c3a85120c7dfe94efb69c5cb4db60531ac924289","subject":"Merge \"Update hypervisor support matrix with kvm on system z\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3a85120c7dfe94efb69c5cb4db60531ac924289"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-05 13:31:55.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-05 13:31:55.000000000","tz":120},"subject":"Remove memory validation for live migration","message":"Remove memory validation for live migration\n\nSince live migration has been moved to the conductor, there is no possibility\nfor the conductor to verify if the destination has enough RAM just because it\ndoesn\u0027t know the allocation ratios given by the scheduler.\n\nAs it would be a huge layer separation violation to import the ratio opts in the\nconductor, we prefer to remove the check. To be honest, if the operator provides\na target, that\u0027s his duty to make sure the rules are enforced, since we are not\nchecking all the scheduler logics as it\u0027s just bypassed.\n\nA separate blueprint check-destination-on-migrations is proposing to optionally\ncheck the scheduler so it will also help this change but for the moment, let\u0027s\nfix the problem which can block a migration.\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\nCloses-Bug: #1451831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f080d42b9617e22f17146b27655111c8759c9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f080d42b9617e22f17146b27655111c8759c9d4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c3a85120c7dfe94efb69c5cb4db60531ac924289","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"175e0e7a30a596379f3098daa1d427980d9b74b3":{"kind":"REWORK","_number":2,"created":"2015-05-06 13:45:07.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/2"}}},"commit":{"parents":[{"commit":"c3a85120c7dfe94efb69c5cb4db60531ac924289","subject":"Merge \"Update hypervisor support matrix with kvm on system z\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3a85120c7dfe94efb69c5cb4db60531ac924289"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-05 13:31:55.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-06 13:44:31.000000000","tz":120},"subject":"Remove memory validation for live migration","message":"Remove memory validation for live migration\n\nSince live migration has been moved to the conductor, there is no possibility\nfor the conductor to verify if the destination has enough RAM just because it\ndoesn\u0027t know the allocation ratios given by the scheduler.\n\nAs it would be a huge layer separation violation to import the ratio opts in the\nconductor, we prefer to remove the check. To be honest, if the operator provides\na target, that\u0027s his duty to make sure the rules are enforced, since we are not\nchecking all the scheduler logics as it\u0027s just bypassed.\n\nA separate blueprint check-destination-on-migrations is proposing to optionally\ncheck the scheduler so it will also help this change but for the moment, let\u0027s\nfix the problem which can block a migration.\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\nCloses-Bug: #1451831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/175e0e7a30a596379f3098daa1d427980d9b74b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/175e0e7a30a596379f3098daa1d427980d9b74b3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c3a85120c7dfe94efb69c5cb4db60531ac924289","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"30d05ee5e33a29967f9409503b0c6da60dca1335":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2015-05-26 13:08:06.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/3"}}},"commit":{"parents":[{"commit":"b6e5d68c666a1fa2f0dd90de47cd7ec0a2e32438","subject":"Merge \"Disable oslo.vmware test dependency on Python 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6e5d68c666a1fa2f0dd90de47cd7ec0a2e32438"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-05 13:31:55.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-26 13:07:33.000000000","tz":120},"subject":"Remove memory validation for live migration","message":"Remove memory validation for live migration\n\nSince live migration has been moved to the conductor, there is no possibility\nfor the conductor to verify if the destination has enough RAM just because it\ndoesn\u0027t know the allocation ratios given by the scheduler.\n\nAs it would be a huge layer separation violation to import the ratio opts in the\nconductor, we prefer to remove the check. To be honest, if the operator provides\na target, that\u0027s his duty to make sure the rules are enforced, since we are not\nchecking all the scheduler logics as it\u0027s just bypassed.\n\nA separate blueprint check-destination-on-migrations is proposing to optionally\ncheck the scheduler so it will also help this change but for the moment, let\u0027s\nfix the problem which can block a migration.\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\nCloses-Bug: #1451831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30d05ee5e33a29967f9409503b0c6da60dca1335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30d05ee5e33a29967f9409503b0c6da60dca1335"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b6e5d68c666a1fa2f0dd90de47cd7ec0a2e32438","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"70f5c501b51b23ce11f936de0b6fd2f7278df224":{"kind":"REWORK","_number":4,"created":"2015-09-17 15:23:53.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/4"}}},"commit":{"parents":[{"commit":"5d34c572cb4b04e4df1fa1103b456787972c6332","subject":"Merge \"Change v3 to v21 for devref api_plugins.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d34c572cb4b04e4df1fa1103b456787972c6332"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-05-05 13:31:55.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-09-17 15:21:59.000000000","tz":120},"subject":"Correct memory validation for live migration","message":"Correct memory validation for live migration\n\nSince live migration has been moved to the conductor, there was no possibility\nfor the conductor to verify if the destination had enough RAM just because it\ndidn\u0027t know the allocation ratios given by the scheduler.\n\nNow that ComputeNodes provide a ram_allocation_field, we can fix that check and\nprovide the same validation than RAMFilter to make sure that the destination\nis good.\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\nCloses-Bug: #1451831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70f5c501b51b23ce11f936de0b6fd2f7278df224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70f5c501b51b23ce11f936de0b6fd2f7278df224"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5d34c572cb4b04e4df1fa1103b456787972c6332","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"81ecf2aab9cd1d2a5c7d3d883d3093231d0755db":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2015-09-17 15:52:36.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/5"}}},"commit":{"parents":[{"commit":"5d34c572cb4b04e4df1fa1103b456787972c6332","subject":"Merge \"Change v3 to v21 for devref api_plugins.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d34c572cb4b04e4df1fa1103b456787972c6332"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-09-17 15:51:06.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-09-17 15:52:11.000000000","tz":120},"subject":"Correct memory validation for live migration","message":"Correct memory validation for live migration\n\nSince live migration has been moved to the conductor, there was no\npossibility for the conductor to verify if the destination had\nenough RAM just because it didn\u0027t know the allocation ratios given\nby the scheduler.\n\nNow that ComputeNodes provide a ram_allocation_field, we can fix\nthat check and provide the same validation than RAMFilter to make\nsure that the destination is good.\n\nCloses-Bug: #1451831\nCloses-Bug: #1214943\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81ecf2aab9cd1d2a5c7d3d883d3093231d0755db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81ecf2aab9cd1d2a5c7d3d883d3093231d0755db"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5d34c572cb4b04e4df1fa1103b456787972c6332","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"370acf4a8581deb0555395f7d24d7306d64c84b1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2015-09-21 14:18:48.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/6"}}},"commit":{"parents":[{"commit":"bbaa721e4f401b6895aa7456ddca45502b32df11","subject":"Merge \"libvirt: fix direct OVS plugging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bbaa721e4f401b6895aa7456ddca45502b32df11"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-09-17 15:51:06.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-09-21 14:18:29.000000000","tz":120},"subject":"Correct memory validation for live migration","message":"Correct memory validation for live migration\n\nSince live migration has been moved to the conductor, there was no\npossibility for the conductor to verify if the destination had\nenough RAM just because it didn\u0027t know the allocation ratios given\nby the scheduler.\n\nNow that ComputeNodes provide a ram_allocation_field, we can fix\nthat check and provide the same validation than RAMFilter to make\nsure that the destination is good.\n\nCloses-Bug: #1451831\nCloses-Bug: #1214943\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/370acf4a8581deb0555395f7d24d7306d64c84b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/370acf4a8581deb0555395f7d24d7306d64c84b1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bbaa721e4f401b6895aa7456ddca45502b32df11","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f5e35dcfe8ee586106438bcfa551426babc75bf6":{"kind":"REWORK","_number":7,"created":"2015-10-05 13:35:06.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/51/180151/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/180151/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/180151/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/180151/7"}}},"commit":{"parents":[{"commit":"bbaa721e4f401b6895aa7456ddca45502b32df11","subject":"Merge \"libvirt: fix direct OVS plugging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bbaa721e4f401b6895aa7456ddca45502b32df11"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-09-17 15:51:06.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2015-10-05 13:34:21.000000000","tz":120},"subject":"Correct memory validation for live migration","message":"Correct memory validation for live migration\n\nSince live migration has been moved to the conductor, there was no\npossibility for the conductor to verify if the destination had\nenough RAM just because it didn\u0027t know the allocation ratios given\nby the scheduler.\n\nNow that ComputeNodes provide a ram_allocation_field, we can fix\nthat check and provide the same validation than RAMFilter to make\nsure that the destination is good.\n\nCloses-Bug: #1451831\nCloses-Bug: #1214943\n\nChange-Id: Ie6c768fc915553da73160ea51961078bfbacec77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5e35dcfe8ee586106438bcfa551426babc75bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5e35dcfe8ee586106438bcfa551426babc75bf6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bbaa721e4f401b6895aa7456ddca45502b32df11","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
