)]}'
{"id":"openstack%2Fnova~187857","triplet_id":"openstack%2Fnova~master~I28ba1970317f3d5691d05283796d4345ff38a804","project":"openstack/nova","branch":"master","topic":"bug/1369465","hashtags":[],"change_id":"I28ba1970317f3d5691d05283796d4345ff38a804","subject":"libvirt: Split out resize_image logic from create_image","status":"MERGED","created":"2015-06-03 04:08:55.000000000","updated":"2016-08-18 17:48:29.000000000","submitted":"2015-09-28 17:49:10.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11e6579c559eeeb0188720955f741ae9a2fff33b","_number":187857,"virtual_id_number":187857,"owner":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2015-09-15 03:16:45.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-09-15 03:05:52.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":2,"date":"2015-09-28 17:49:10.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-09-15 02:31:14.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2015-09-15 02:00:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},{"value":0,"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},{"value":1,"date":"2015-09-15 02:18:52.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2016-08-16 14:21:59.000000000","post_submit":true,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},{"value":0,"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2015-09-15 01:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":1,"date":"2015-09-15 04:36:33.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"date":"2015-09-15 04:06:36.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-09-28 14:52:07.000000000","_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},{"value":0,"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":1,"date":"2015-09-24 09:23:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},{"value":1,"date":"2015-09-15 01:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2015-09-23 01:13:33.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":1,"date":"2015-09-25 06:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2015-09-25 13:40:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-09-28 14:52:07.000000000","_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},{"value":0,"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},{"value":0,"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},{"value":0,"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},{"value":0,"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"value":0,"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},{"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},{"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},{"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},{"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-06-08 09:06:02.000000000","updated_by":{"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},"reviewer":{"_account_id":8587,"name":"Li Min Liu","email":"liulm@certusnet.com.cn","username":"LarryLiu"},"state":"REVIEWER"},{"updated":"2015-07-28 19:45:32.000000000","updated_by":{"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},"reviewer":{"_account_id":6984,"name":"Sébastien Han","email":"seb@redhat.com","username":"leseb"},"state":"REVIEWER"},{"updated":"2015-07-28 21:26:37.000000000","updated_by":{"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},"reviewer":{"_account_id":8582,"name":"Zoltan Arnold Nagy","email":"nag@zurich.ibm.com","username":"zoltan"},"state":"REVIEWER"},{"updated":"2015-08-22 11:10:02.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2015-09-09 06:15:16.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2015-09-09 18:15:02.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-09 20:54:08.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-09 23:23:16.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-09-14 19:17:14.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2015-09-14 19:24:38.000000000","updated_by":{"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},"reviewer":{"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},"state":"REVIEWER"},{"updated":"2015-09-15 01:38:51.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 02:00:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 02:18:52.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 02:31:14.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 03:05:52.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 03:16:45.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 04:06:36.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-15 04:36:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-09-23 01:13:33.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2015-09-24 09:23:47.000000000","updated_by":{"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},"reviewer":{"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},"state":"REVIEWER"},{"updated":"2015-09-25 06:45:52.000000000","updated_by":{"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},"reviewer":{"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},"state":"REVIEWER"},{"updated":"2015-09-25 13:40:24.000000000","updated_by":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"reviewer":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"state":"REVIEWER"},{"updated":"2015-09-28 14:52:07.000000000","updated_by":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"reviewer":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"state":"REVIEWER"},{"updated":"2015-09-28 17:49:10.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-08-16 14:21:59.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"f826e36ace7cc1435e7bbf235b48887362f3bafc","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-03 04:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f16b705b11c5beb5c5703db39c22a5043376031","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-03 04:09:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"ec155b32c09a62c0ecf1dcb8507e854c23824cdd","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-03 04:10:01.000000000","message":"Patch Set 1:\n\n* pci-test http://198.175.100.33/187857/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a78329346af6a9a4cf4ebb84479d3535cba30d9b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-03 04:25:44.000000000","message":"Patch Set 2:\n\n* pci-test http://198.175.100.33/187857/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8379abd2cdf3e0be998efe02525e8929c4e1e257","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-03 04:32:31.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/049f48cc41cd448da569c5ef9934799c/index.html : SUCCESS in 20m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/45b968780e214c15845a8f0afa33a310/index.html : SUCCESS in 21m 05s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5e40623f031d4b9ea2275c26015c3e67/index.html : SUCCESS in 17m 30s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/2/check/real-db-upgrade_nova_percona_user_001:th-percona/d04a8ca96b6749c9a9069c16a9aec4ec/index.html : SUCCESS in 21m 52s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/457a145b20c94ff393c257960878fe71/index.html : SUCCESS in 18m 12s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/2/check/real-db-upgrade_nova_percona_user_002:th-percona/7c3d18c15e124ea68daafb75452a2df1/index.html : SUCCESS in 21m 08s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"04c95f3c5e6612cc86729613a3608d843f288f25","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-03 04:47:08.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/2/ : FAILURE in 36m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8599e11a4af0dd226d3dccac100e2902f1ae9fd1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-03 05:20:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/2/check/dsvm-tempest-xen/7e8903b : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":2},{"id":"cb48a03a7ab4ac5eb0460ab1c48e9c1b7d81010f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-03 05:21:35.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/2/check-ibm-tempest-dsvm-full/5fee58e : SUCCESS in 48m 53s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/2/check-ibm-tempest-dsvm-postgres-full/b08b453 : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9bf8636472260d5414c5b22b96af0dc697ff0d84","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-03 05:23:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/2/check/gate-nova-pep8/9fac634/ : SUCCESS in 14m 47s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/2/check/gate-nova-docs/a390bb9//doc/build/html/ : SUCCESS in 8m 17s\n- gate-nova-python27 http://logs.openstack.org/57/187857/2/check/gate-nova-python27/a87a7b4/ : SUCCESS in 13m 06s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-full/62a5504/ : SUCCESS in 41m 28s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-postgres-full/3e86074/ : SUCCESS in 32m 36s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-neutron-full/75755c3/ : SUCCESS in 44m 03s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/2/check/check-grenade-dsvm/b8a138e/ : SUCCESS in 36m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/2/check/gate-tempest-dsvm-large-ops/72f59a2/ : SUCCESS in 22m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/2/check/gate-tempest-dsvm-neutron-large-ops/dc389c7/ : SUCCESS in 21m 38s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/2/check/check-devstack-dsvm-cells/b465556/ : SUCCESS in 17m 05s\n- gate-nova-python34 http://logs.openstack.org/57/187857/2/check/gate-nova-python34/c724499/ : SUCCESS in 10m 49s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/2/check/gate-nova-tox-functional/a58b02a/ : SUCCESS in 11m 17s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/2/check/check-grenade-dsvm-partial-ncpu/b37d8aa/ : SUCCESS in 36m 09s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-ironic-pxe_ssh/30d73ab/ : SUCCESS in 33m 48s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-nova-v21-full/7852386/ : SUCCESS in 33m 34s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-cells/c1da2da/ : SUCCESS in 30m 41s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-full-ceph/296b7dd/ : SUCCESS in 34m 10s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/2/check/check-tempest-dsvm-multinode-full/a9683f6/ : SUCCESS in 50m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"209af33eae8a67ad035aefdfb47282932e0e482a","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-03 18:53:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"945fb02864b683bf83f2bc419eb92add17416b1a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-03 19:36:53.000000000","message":"Patch Set 3:\n\n* pci-test http://198.175.100.33/187857/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5c15046719c53f1f97041ad1e1ae13fe5660f24b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-03 19:52:45.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6e54ee6c10cc40eea5c80575496a9062/index.html : SUCCESS in 20m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/020d92e05bd24237994eab88760a4de9/index.html : SUCCESS in 25m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ef07228e6f39459db6a9f5204de5c22e/index.html : SUCCESS in 20m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/3/check/real-db-upgrade_nova_percona_user_001:th-percona/6b19b2689f7e4900a8f39e6608b7fb52/index.html : SUCCESS in 22m 12s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/fa8900a3b4a54920a5828716a4070802/index.html : SUCCESS in 21m 09s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/3/check/real-db-upgrade_nova_percona_user_002:th-percona/fb1e8efa434b4ce1be864c39c3777aba/index.html : SUCCESS in 17m 12s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"d41db13779bca5aeba4e94095b3c6ef25377926f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-03 20:05:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/3/check/dsvm-tempest-xen/b87ec45 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":3},{"id":"9ea2cf4275f42c1354adf1f09e91f30214a7435b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-03 20:24:20.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/3/check-ibm-tempest-dsvm-full/71548c4 : SUCCESS in 1h 04m 35s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/3/check-ibm-tempest-dsvm-postgres-full/9132226 : SUCCESS in 1h 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c042fb7221f973ee4105bd51d2565febd23062a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-03 21:08:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/187857/3/ : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":3},{"id":"633a27c732c9b842828c9504adf9c79ddceeceea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-03 21:27:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/3/check/gate-nova-pep8/5fbaa69/ : SUCCESS in 6m 42s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/3/check/gate-nova-docs/e04a26d//doc/build/html/ : SUCCESS in 8m 11s\n- gate-nova-python27 http://logs.openstack.org/57/187857/3/check/gate-nova-python27/9631380/ : SUCCESS in 11m 01s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-full/38a4358/ : SUCCESS in 40m 05s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-postgres-full/9bfcc95/ : SUCCESS in 40m 25s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-neutron-full/2477743/ : SUCCESS in 54m 57s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/3/check/check-grenade-dsvm/bb39725/ : SUCCESS in 40m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/3/check/gate-tempest-dsvm-large-ops/21a50c4/ : SUCCESS in 26m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/3/check/gate-tempest-dsvm-neutron-large-ops/b245a4b/ : SUCCESS in 23m 38s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/3/check/check-devstack-dsvm-cells/7e101f5/ : SUCCESS in 19m 23s\n- gate-nova-python34 http://logs.openstack.org/57/187857/3/check/gate-nova-python34/28eab11/ : SUCCESS in 13m 07s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/3/check/gate-nova-tox-functional/38b5ee0/ : SUCCESS in 9m 18s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/3/check/check-grenade-dsvm-partial-ncpu/d75ca3a/ : SUCCESS in 40m 21s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-ironic-pxe_ssh/a9ce864/ : SUCCESS in 33m 07s\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-nova-v21-full/2ffb860/ : SUCCESS in 35m 32s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-cells/52ca3ee/ : SUCCESS in 37m 33s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-full-ceph/10b4101/ : SUCCESS in 40m 21s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/3/check/check-tempest-dsvm-multinode-full/cd5bde5/ : FAILURE in 2h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af57c4f68d56d2b58a2cd94b5c1c90adec3888c0","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-08 22:44:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e64839c9ffa83e8c3015895d22124c2744e9a4df","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-06-08 23:00:56.000000000","message":"Patch Set 4:\n\n* pci-test http://198.175.100.33/187857/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"62a440b83ce203989b70518bc362b21e4d9d1214","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-08 23:11:26.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9f92e071f0684a2782c05c1dcabeee07/index.html : SUCCESS in 19m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/de67ccf61c86474dba805265bdedaf31/index.html : SUCCESS in 26m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dc3844fd733449e3a119c67a7127d4f3/index.html : SUCCESS in 19m 46s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/4/check/real-db-upgrade_nova_percona_user_001:th-percona/445b86e3193340a6b17d5f6c97949ffc/index.html : SUCCESS in 21m 41s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/fa264a95dac247bc8b123a89f056d1dc/index.html : SUCCESS in 20m 44s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/4/check/real-db-upgrade_nova_percona_user_002:th-percona/234ee6e0f1d141c989cc0b1d75a9baea/index.html : SUCCESS in 20m 23s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"e61dc0a09a6b289ef7c6d7361a17d3b25b8c5613","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-08 23:28:30.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/4/ : FAILURE in 44m 02s","accounts_in_message":[],"_revision_number":4},{"id":"b75cb3d79d08cd67434a66a1f74e9c96dbd0d56a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-08 23:48:10.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/4/368/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"2ffd2cbaabb7dc7fdf95a619e5f34eeb973f2900","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-08 23:50:21.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/4/check-ibm-tempest-dsvm-full/36bcfab : SUCCESS in 1h 05m 59s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/4/check-ibm-tempest-dsvm-postgres-full/983bccd : SUCCESS in 1h 00m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"47cd92027c70ca2241376b0c4a4a91d3e3d4606c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-09 00:04:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/4/check/dsvm-tempest-xen/75a9757 : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":4},{"id":"71bd48dfc91574669cd65b71eeab49f320ea8dce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-09 00:19:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/4/check/gate-nova-pep8/1306484/ : SUCCESS in 7m 30s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/4/check/gate-nova-docs/f2e9c7d//doc/build/html/ : SUCCESS in 6m 45s\n- gate-nova-python27 http://logs.openstack.org/57/187857/4/check/gate-nova-python27/7ef158c/ : SUCCESS in 13m 32s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-full/25856e5/ : SUCCESS in 45m 35s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-postgres-full/42bf238/ : SUCCESS in 44m 48s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-neutron-full/e39db4a/ : SUCCESS in 1h 01m 35s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/4/check/check-grenade-dsvm/177ef09/ : SUCCESS in 39m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/4/check/gate-tempest-dsvm-large-ops/59a7a5c/ : SUCCESS in 24m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/4/check/gate-tempest-dsvm-neutron-large-ops/f00a944/ : SUCCESS in 26m 00s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/4/check/check-devstack-dsvm-cells/3a24ea8/ : SUCCESS in 21m 22s\n- gate-nova-python34 http://logs.openstack.org/57/187857/4/check/gate-nova-python34/ab9fddc/ : SUCCESS in 9m 34s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/4/check/gate-nova-tox-functional/c8257fe/ : SUCCESS in 13m 11s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/4/check/check-grenade-dsvm-partial-ncpu/06baee2/ : SUCCESS in 38m 19s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-ironic-pxe_ssh/65943b6/ : SUCCESS in 35m 21s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-docker/4b3d60c/ : SUCCESS in 36m 43s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-nova-v21-full/19baaa5/ : SUCCESS in 34m 56s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-cells/e403bad/ : FAILURE in 32m 03s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-full-ceph/9447857/ : SUCCESS in 36m 40s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/4/check/check-tempest-dsvm-multinode-full/5153735/ : FAILURE in 1h 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a5ac8e390b26eda85e6a5f050fd5d3fcdff64159","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-06-16 03:38:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/4 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d39dd73d9145ca36e766d80591ba892b7726dd00","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-16 17:10:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"830c269cc778d3804213a06ba770fcf99da3a38e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-16 17:27:44.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/5/check-ibm-tempest-dsvm-full/107a840 : FAILURE in 13m 00s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/5/check-ibm-tempest-dsvm-postgres-full/f026f72 : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"38dfe95e102b556e74b06807495d32dc42bf3ec6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-16 18:07:30.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/5/1316/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"064d2b6cec9229b281647f50785144bd853b0a87","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-16 18:10:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/5/check/gate-nova-pep8/12f845b/ : SUCCESS in 8m 35s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/5/check/gate-nova-docs/a6f6bf2//doc/build/html/ : SUCCESS in 9m 46s\n- gate-nova-python27 http://logs.openstack.org/57/187857/5/check/gate-nova-python27/4a6732e/ : SUCCESS in 11m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-full/20a074c/ : SUCCESS in 39m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-postgres-full/abfd54a/ : SUCCESS in 37m 20s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-neutron-full/dc41183/ : SUCCESS in 57m 49s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/5/check/check-grenade-dsvm/d348242/ : SUCCESS in 37m 24s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/5/check/gate-tempest-dsvm-large-ops/49623aa/ : SUCCESS in 31m 35s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/5/check/gate-tempest-dsvm-neutron-large-ops/6e4e227/ : SUCCESS in 26m 25s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/5/check/check-devstack-dsvm-cells/3ddf7da/ : FAILURE in 19m 55s\n- gate-nova-python34 http://logs.openstack.org/57/187857/5/check/gate-nova-python34/cf2814a/ : SUCCESS in 10m 15s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/5/check/gate-nova-tox-functional/f4e5d41/ : SUCCESS in 10m 10s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/5/check/check-grenade-dsvm-partial-ncpu/eed928a/ : SUCCESS in 45m 10s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-ironic-pxe_ssh/33238d3/ : SUCCESS in 32m 43s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-docker/4201f0f/ : FAILURE in 39m 08s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-nova-v21-full/2f45d23/ : SUCCESS in 39m 05s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-cells/d3815cb/ : SUCCESS in 45m 43s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-full-ceph/1538c88/ : SUCCESS in 48m 39s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/5/check/check-tempest-dsvm-multinode-full/01c2987/ : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d905135abf9804728f4975994728a39849e2c3fb","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-16 18:10:54.000000000","message":"Patch Set 5:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":5},{"id":"282bc2d475b4c4a9327d044f1ace9ac63f004dfe","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-16 18:30:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/5/check/dsvm-tempest-xen/5ca5282 : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":5},{"id":"421d42b7550b07df48332baa6ae42a1d31486cb3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-16 20:26:03.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/5/ : FAILURE in 39m 46s","accounts_in_message":[],"_revision_number":5},{"id":"23cae159e742c2303dc30fdf55e27eaf73ed6f7b","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-17 17:51:31.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8716054fc61ec9000da1550358cdb95b2c48cc66","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-17 18:48:29.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/6/check-ibm-tempest-dsvm-full/3b31a4e : SUCCESS in 56m 32s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/6/check-ibm-tempest-dsvm-postgres-full/9865243 : SUCCESS in 56m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f97ccbde66e170bf4d67d758a94d51e3e7d12e66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-17 18:53:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/6/check/gate-nova-pep8/1860b86/ : SUCCESS in 8m 50s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/6/check/gate-nova-docs/59d9cfe//doc/build/html/ : SUCCESS in 7m 31s\n- gate-nova-python27 http://logs.openstack.org/57/187857/6/check/gate-nova-python27/e2698f0/ : SUCCESS in 14m 12s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-full/a03d214/ : SUCCESS in 56m 03s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-postgres-full/9a20643/ : SUCCESS in 51m 33s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-neutron-full/e1ef02e/ : SUCCESS in 56m 19s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/6/check/check-grenade-dsvm/03e05ec/ : SUCCESS in 44m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/6/check/gate-tempest-dsvm-large-ops/741bab8/ : SUCCESS in 31m 37s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/6/check/gate-tempest-dsvm-neutron-large-ops/454ca83/ : SUCCESS in 36m 11s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/6/check/check-devstack-dsvm-cells/3c3fc13/ : FAILURE in 29m 13s\n- gate-nova-python34 http://logs.openstack.org/57/187857/6/check/gate-nova-python34/d4dd20a/ : SUCCESS in 9m 58s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/6/check/gate-nova-tox-functional/a17be59/ : SUCCESS in 12m 30s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/6/check/check-grenade-dsvm-partial-ncpu/2af9d30/ : SUCCESS in 40m 10s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-ironic-pxe_ssh/04a9017/ : SUCCESS in 35m 12s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-docker/08f882e/ : SUCCESS in 40m 52s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-nova-v21-full/9f7fc75/ : SUCCESS in 52m 30s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-cells/bea513a/ : SUCCESS in 45m 58s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-full-ceph/f4249c1/ : SUCCESS in 45m 02s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/6/check/check-tempest-dsvm-multinode-full/c26655a/ : SUCCESS in 46m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37561bdd993253ae73256602755ff61b2aca6107","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-17 18:53:30.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"70c6815be3059f69a0ca7342ec03d86f5c6263e8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-17 19:02:13.000000000","message":"Patch Set 6: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/6/1478/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"799cdda84128083b6de488bedb502a0eafb2d045","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-17 19:05:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/6/check/dsvm-tempest-xen/6911f35 : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":6},{"id":"b5524479fa1fff433a0e65a94768f00cb926b449","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-17 19:56:53.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b8180414b86442d7b387be97c0b89556/index.html : SUCCESS in 20m 32s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ed24264d79e44294a107b16a7371dae8/index.html : SUCCESS in 36m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a6b3392d2bb34f68b50160ab891819aa/index.html : SUCCESS in 18m 36s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/6/check/real-db-upgrade_nova_percona_user_001:th-percona/d58b2c0922dc42c3b3ca71d7d44c8af1/index.html : SUCCESS in 26m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/687d0cbefe55440a956734299fcc8075/index.html : SUCCESS in 18m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/6/check/real-db-upgrade_nova_percona_user_002:th-percona/d3ca2f6962f1462e92d83fb3a79fd1e7/index.html : SUCCESS in 20m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"f1b35ee56dc2c9d9151d475b3fed47c479d55474","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-17 21:34:07.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/6/ : FAILURE in 45m 03s","accounts_in_message":[],"_revision_number":6},{"id":"189e9f7eccbf27fe2d839fa782d91e9c67921d2b","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-18 17:37:06.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4713cf2935d504136374800d604f25cda0c5ca45","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-18 18:27:08.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/7/check-ibm-tempest-dsvm-full/20979b7 : SUCCESS in 49m 39s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/7/check-ibm-tempest-dsvm-postgres-full/b8f22a8 : SUCCESS in 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"44a205b887eaab0a24af891e0eeb1ad77b98f1dd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-18 18:36:07.000000000","message":"Patch Set 7: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/7/1647/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":7},{"id":"80fcf8bb9baa129f381f859e4990ae57351aedfc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-18 18:59:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/7/check/dsvm-tempest-xen/c7de50a : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":7},{"id":"d55303e316faa85faf4426fdab88bb9eb2c0d1d8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-18 19:23:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/7/check/gate-nova-pep8/ba3e463/ : SUCCESS in 6m 50s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/7/check/gate-nova-docs/9fb9dd9//doc/build/html/ : SUCCESS in 6m 45s\n- gate-nova-python27 http://logs.openstack.org/57/187857/7/check/gate-nova-python27/1c2b4cb/ : SUCCESS in 10m 46s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-full/89d5dc2/ : SUCCESS in 49m 19s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-postgres-full/ada14b2/ : SUCCESS in 41m 42s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-neutron-full/cd9c12e/ : SUCCESS in 56m 07s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/7/check/check-grenade-dsvm/d4050ef/ : SUCCESS in 42m 22s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/7/check/gate-tempest-dsvm-large-ops/8cec05c/ : SUCCESS in 21m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/7/check/gate-tempest-dsvm-neutron-large-ops/6902d2f/ : SUCCESS in 26m 17s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/7/check/check-devstack-dsvm-cells/fe711bc/ : FAILURE in 20m 53s\n- gate-nova-python34 http://logs.openstack.org/57/187857/7/check/gate-nova-python34/09cbd34/ : SUCCESS in 8m 20s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/7/check/gate-nova-tox-functional/ed6e849/ : SUCCESS in 11m 59s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/7/check/check-grenade-dsvm-partial-ncpu/3e74a69/ : SUCCESS in 41m 19s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-ironic-pxe_ssh/529c9ae/ : SUCCESS in 34m 38s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-docker/75ce3a1/ : SUCCESS in 39m 41s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-nova-v21-full/e1d6cae/ : SUCCESS in 50m 30s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-cells/bdf1830/ : SUCCESS in 44m 13s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-full-ceph/1e339b6/ : SUCCESS in 51m 07s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/7/check/check-tempest-dsvm-multinode-full/92f0cfd/ : FAILURE in 1h 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ff28b998492de4aa22b2b356058593fc16a366b5","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-18 19:23:55.000000000","message":"Patch Set 7:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":7},{"id":"082434a1f8ded711459175533dfa800855f1bc1b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-06-18 20:04:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/7/Nova-ML2-Sriov : SUCCESS in 28m 40s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"6818777f9afc368a9e2a25062fad52bd1b92e87f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-19 00:04:27.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1e3c45233a004da199231729f22ec373/index.html : SUCCESS in 20m 32s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ccb4c9376b2e49a6a4e76b62f3fb5c75/index.html : SUCCESS in 22m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/62c755418ddb426f814e85e152f40192/index.html : SUCCESS in 17m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/7/check/real-db-upgrade_nova_percona_user_001:th-percona/69e1d0f06e354b0f85486f7409122938/index.html : SUCCESS in 24m 34s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7aee9c78020247ff91c0de6b088f32d8/index.html : SUCCESS in 21m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/7/check/real-db-upgrade_nova_percona_user_002:th-percona/41c44762ef294fe9891ef41a5e2b31da/index.html : SUCCESS in 17m 55s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"7f1093be169cfeab7276e3b17aef4b7d95c6785f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-19 00:24:48.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/7/ : FAILURE in 1h 44m 58s","accounts_in_message":[],"_revision_number":7},{"id":"94ceebe2485562713f9baaee77dc8c3b45c6099e","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-19 17:42:08.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"befec693d657350e3b61de2b29fb3e4e604dad64","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-19 18:10:55.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/77b8dfa90a4a46019945e169f83fd7b9/index.html : SUCCESS in 20m 53s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d8649f66a9ef43a6a420831debd80d1d/index.html : SUCCESS in 22m 38s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c83a1541c46c4e14b5c1cafa830ac2a5/index.html : SUCCESS in 20m 16s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/8/check/real-db-upgrade_nova_percona_user_001:th-percona/8304da836b12427a8436a3ebeed6b6f5/index.html : SUCCESS in 27m 48s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/293b84b5727c4a329b092b866add8dd4/index.html : SUCCESS in 20m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/8/check/real-db-upgrade_nova_percona_user_002:th-percona/4eca4735d79e42f19a2c9af355b89a17/index.html : SUCCESS in 17m 54s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"9c72922b28e2cffa73a8688e7e348b650fe996dc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-19 18:47:45.000000000","message":"Patch Set 8: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/8/1765/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":8},{"id":"6f306f2636f78c201c588299bc449b6507cf8b30","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-19 18:47:49.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/8/check-ibm-tempest-dsvm-full/8996985 : SUCCESS in 1h 05m 16s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/8/check-ibm-tempest-dsvm-postgres-full/eadc759 : SUCCESS in 1h 00m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2dac677aa173f9843d1006d120fbbac14d626b64","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-19 18:55:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/8/check/dsvm-tempest-xen/fe529b1 : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":8},{"id":"7d6c1b8b9d6e0f51aa53fba0645fed3f2fb7a921","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-19 19:15:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/8/check/gate-nova-pep8/4d49944/ : SUCCESS in 11m 23s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/8/check/gate-nova-docs/099af51//doc/build/html/ : SUCCESS in 8m 33s\n- gate-nova-python27 http://logs.openstack.org/57/187857/8/check/gate-nova-python27/fc3a5c0/ : SUCCESS in 13m 34s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-full/9933e9e/ : SUCCESS in 50m 10s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-postgres-full/8526968/ : FAILURE in 35m 23s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-neutron-full/2a72e37/ : SUCCESS in 55m 18s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/8/check/check-grenade-dsvm/0b96a6b/ : SUCCESS in 39m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/8/check/gate-tempest-dsvm-large-ops/4274788/ : SUCCESS in 30m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/8/check/gate-tempest-dsvm-neutron-large-ops/27e0a91/ : SUCCESS in 24m 57s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/8/check/check-devstack-dsvm-cells/c455bbb/ : FAILURE in 27m 55s\n- gate-nova-python34 http://logs.openstack.org/57/187857/8/check/gate-nova-python34/6bdb5ed/ : SUCCESS in 9m 44s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/8/check/gate-nova-tox-functional/429e6fb/ : SUCCESS in 10m 12s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/8/check/check-grenade-dsvm-partial-ncpu/129bee5/ : SUCCESS in 42m 00s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-ironic-pxe_ssh/097beb8/ : SUCCESS in 37m 54s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-docker/382a417/ : SUCCESS in 40m 05s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-nova-v21-full/8a8d092/ : SUCCESS in 44m 12s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-cells/ef66b33/ : SUCCESS in 35m 59s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-full-ceph/e70578e/ : SUCCESS in 45m 54s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/8/check/check-tempest-dsvm-multinode-full/a41ef0c/ : FAILURE in 1h 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"468f7ad8090422d02253950b21fb9ab39c87370b","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-19 19:15:32.000000000","message":"Patch Set 8:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1353962\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":8},{"id":"9b133eef5926fdcdc74656e8498069defc49fcef","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-06-19 21:05:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/8/Nova-ML2-Sriov : SUCCESS in 2h 08m 16s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"c41b19bef4592575acec591456e6e1674ea1cd58","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-19 22:52:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/187857/8/ : SUCCESS in 1h 29m 45s","accounts_in_message":[],"_revision_number":8},{"id":"fd138dbce1fd639801da5cf5a5ff384bd301f3a5","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-06-22 16:51:31.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"f311f648496242f9a1ef0bc3bed9b5b89b173416","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-06-22 17:35:16.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/9/Nova-ML2-Sriov : SUCCESS in 23m 36s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"ac816c7605b8d1d02be4c9f122b0a070e6d701d2","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-22 17:42:38.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8f7dfac1a87c47d2888e219dc462d438/index.html : SUCCESS in 21m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/90b20a24797c45f1b67dff7d715f3951/index.html : SUCCESS in 28m 49s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fda5dc5c107d4c7eb1be79632a126963/index.html : SUCCESS in 21m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/9/check/real-db-upgrade_nova_percona_user_001:th-percona/7871fe30c03e4a53adb3a7b714420508/index.html : SUCCESS in 23m 01s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c0fd3d63ad354d17aafc6303772dabd9/index.html : SUCCESS in 22m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/9/check/real-db-upgrade_nova_percona_user_002:th-percona/84e59ee97cf44dc797476d217626de1f/index.html : SUCCESS in 21m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"bba9e024f9fed7ab30ac2582b9b355aff6d8d01a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-22 17:50:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/9/check/gate-nova-pep8/6b0457e/ : SUCCESS in 6m 23s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/9/check/gate-nova-docs/37a095a//doc/build/html/ : SUCCESS in 6m 21s\n- gate-nova-python27 http://logs.openstack.org/57/187857/9/check/gate-nova-python27/017a69b/ : FAILURE in 13m 06s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-full/a9cf843/ : SUCCESS in 51m 28s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-postgres-full/8cd6347/ : SUCCESS in 51m 31s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-neutron-full/dab39f0/ : SUCCESS in 56m 33s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/9/check/check-grenade-dsvm/339702b/ : SUCCESS in 38m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/9/check/gate-tempest-dsvm-large-ops/695a9dc/ : SUCCESS in 25m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/9/check/gate-tempest-dsvm-neutron-large-ops/c28bfbf/ : SUCCESS in 20m 57s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/9/check/check-devstack-dsvm-cells/4120083/ : FAILURE in 27m 27s\n- gate-nova-python34 http://logs.openstack.org/57/187857/9/check/gate-nova-python34/37b35f7/ : SUCCESS in 10m 58s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/9/check/gate-nova-tox-functional/7d86497/ : SUCCESS in 9m 38s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/9/check/check-grenade-dsvm-partial-ncpu/ffc459b/ : SUCCESS in 36m 19s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-ironic-pxe_ssh/b695ad9/ : SUCCESS in 39m 31s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-docker/9ff0761/ : SUCCESS in 37m 01s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-nova-v21-full/06ff921/ : SUCCESS in 41m 32s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-cells/1df55ca/ : SUCCESS in 40m 06s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-full-ceph/f322be6/ : SUCCESS in 40m 45s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/9/check/check-tempest-dsvm-multinode-full/018a23d/ : SUCCESS in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4766759b9d3389cb99dadd77d086ff94ed476bd6","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-22 17:50:32.000000000","message":"Patch Set 9:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- gate-nova-python27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"990e0114ad0eaf21e7ec1d74aa2016570fcc229d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-22 17:56:45.000000000","message":"Patch Set 9: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/9/1915/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":9},{"id":"d9145f2a3fd27db11db8d195d6f510d36039d7c9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-22 18:11:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/9/check/dsvm-tempest-xen/eb340da : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":9},{"id":"840e6d11c4edd7eb4e344b162208f4db10632d33","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-22 18:12:39.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/9/check-ibm-tempest-dsvm-full/b75b579 : SUCCESS in 1h 04m 42s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/9/check-ibm-tempest-dsvm-postgres-full/575e022 : SUCCESS in 46m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4423ac6eaed2a1217aec1e676350f1fe13b704af","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-22 19:16:56.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/9/ : FAILURE in 1h 54m 22s","accounts_in_message":[],"_revision_number":9},{"id":"1b1152e24990a26397987a65aa5e662256f30aca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-06-24 19:19:00.000000000","message":"Patch Set 9:\n\nBuild failed. Logs: http://208.91.1.172/logs/187857/9 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"6244518ddb19843ae9d22802f4b0e952726a4225","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-08 21:08:31.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"9f0a6a90555332b24f42c60d1eb0d4cd45cc1df3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-08 22:08:49.000000000","message":"Patch Set 10: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/10/3759/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":10},{"id":"7e14cd4a283ca109347e29d6142790c896a9009c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-08 22:35:55.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/10/check-ibm-tempest-dsvm-full/cb8f272 : SUCCESS in 51m 43s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/10/check-ibm-tempest-dsvm-postgres-full/901eef7 : SUCCESS in 1h 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ab9058e621c96a730b9745570531dbd9d0b466a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-08 22:44:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/10/check/gate-nova-pep8/ffff42e/ : SUCCESS in 7m 06s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/10/check/gate-nova-docs/7969da5//doc/build/html/ : SUCCESS in 6m 46s\n- gate-nova-python27 http://logs.openstack.org/57/187857/10/check/gate-nova-python27/0d349ef/ : FAILURE in 10m 55s\n- gate-nova-python34 http://logs.openstack.org/57/187857/10/check/gate-nova-python34/d755af8/ : SUCCESS in 9m 11s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-full/981523f/ : SUCCESS in 44m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-postgres-full/4993e91/ : SUCCESS in 56m 11s\n- check-tempest-dsvm-neutron http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-neutron/4ce2bb7/ : SUCCESS in 42m 55s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-neutron-full/d8f83ee/ : SUCCESS in 49m 12s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/10/check/check-grenade-dsvm/9581499/ : FAILURE in 34m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/10/check/gate-tempest-dsvm-large-ops/800e5f6/ : SUCCESS in 36m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/10/check/gate-tempest-dsvm-neutron-large-ops/a5dc97a/ : SUCCESS in 30m 09s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/10/check/check-devstack-dsvm-cells/418f7b0/ : FAILURE in 26m 44s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/10/check/gate-nova-tox-functional/1ce7745/ : SUCCESS in 9m 50s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/10/check/check-grenade-dsvm-partial-ncpu/bc96b8e/ : SUCCESS in 50m 33s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-ironic-pxe_ssh/a569b17/ : SUCCESS in 43m 30s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-docker/57090cf/ : SUCCESS in 44m 34s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-nova-v21-full/53e56a2/ : FAILURE in 56m 15s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-cells/d8aebcd/ : SUCCESS in 46m 47s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-full-ceph/317c960/ : SUCCESS in 57m 11s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-multinode-full/4531d47/ : FAILURE in 1h 22m 54s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-multinode-smoke/2c78bf0/ : FAILURE in 47m 02s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/10/check/check-tempest-dsvm-neutron-multinode-smoke/39d0783/ : FAILURE in 59m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f3f5fbb56115390a1ad49550cdce46812e9218d3","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-08 22:44:39.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1446583\n- check-tempest-dsvm-nova-v21-full: https://bugs.launchpad.net/bugs/1355573\n- gate-nova-python27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"145a1ccbda5dd8d80cac8ec28068c5f0cdbb961a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-08 22:58:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/10/check/dsvm-tempest-xen/e0ff8ba : FAILURE in 1h 16m 50s","accounts_in_message":[],"_revision_number":10},{"id":"4f3ff5d8fe332ce005fd166b328d091dbd88de27","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-08 23:36:42.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/10/ : FAILURE in 45m 47s","accounts_in_message":[],"_revision_number":10},{"id":"ae0522fe6f7088296e987eadfa5898a901424920","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-07-09 06:44:38.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nIt may be worthwhile basing this on https://review.openstack.org/#/c/193195 series...","accounts_in_message":[],"_revision_number":10},{"id":"8dcf8307c062d2c05748c67894f25c703e313a2c","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-09 18:49:39.000000000","message":"Patch Set 11: Commit message was updated","accounts_in_message":[],"_revision_number":11},{"id":"f65dae02a41350aaac3f0725d446726c8d427d4b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-09 19:20:22.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5a19b66f7bd14d6ab095db0894dc91b1/index.html : SUCCESS in 20m 28s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/02fc899cabf04271b81e7b8d916ca6a9/index.html : SUCCESS in 27m 35s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0055440740ea4c1a91659b46efbaa413/index.html : SUCCESS in 20m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/11/check/real-db-upgrade_nova_percona_user_001:th-percona/a18c7c944ee549aa8ea44da7aba4e25e/index.html : SUCCESS in 29m 05s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/19d95a6862f54a7c966052a5c152329d/index.html : SUCCESS in 21m 12s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/11/check/real-db-upgrade_nova_percona_user_002:th-percona/26c0fc69098740b3bc4145fd9b533163/index.html : SUCCESS in 21m 11s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"9d61087e6ee9466966a7d6fc98102806a64efafd","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-09 19:26:04.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5b930fb42dac0ff0a5c09b9d6e6ecbb58f2e5b4c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-09 19:54:45.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7fd0bd29709940b88849c44bbe3f80be/index.html : SUCCESS in 18m 05s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5401d69af493407ab77882e78db756bf/index.html : SUCCESS in 27m 35s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dcfb7a2427f14d8c819612063e3505c4/index.html : SUCCESS in 21m 09s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/12/check/real-db-upgrade_nova_percona_user_001:th-percona/77d8d164d40c4bcb8f436244f4524f14/index.html : SUCCESS in 27m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/44f1030b679d4df592b8befa576460ef/index.html : SUCCESS in 18m 11s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/12/check/real-db-upgrade_nova_percona_user_002:th-percona/304fa508fe744b04adaff5d0c9979b48/index.html : SUCCESS in 21m 07s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"c9c5b5d2a7ce92c0ab1efb9c4c94e238829198b9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-09 20:25:07.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/12/check-ibm-tempest-dsvm-full/5bd4870 : SUCCESS in 58m 32s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/12/check-ibm-tempest-dsvm-postgres-full/e83fd62 : SUCCESS in 51m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6ffafbd0a504c22fd7660fdadd8da6f4a2e8b76b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-09 20:28:41.000000000","message":"Patch Set 12: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/12/3859/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":12},{"id":"910375775b56ef25a2d3d822daa7bd3cff682e81","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-09 20:30:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/12/check/gate-nova-pep8/0b49f9b/ : SUCCESS in 7m 18s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/12/check/gate-nova-docs/73d4bfb//doc/build/html/ : SUCCESS in 6m 46s\n- gate-nova-python27 http://logs.openstack.org/57/187857/12/check/gate-nova-python27/0fdaf76/ : SUCCESS in 10m 40s\n- gate-nova-python34 http://logs.openstack.org/57/187857/12/check/gate-nova-python34/299243a/ : SUCCESS in 10m 17s\n- check-tempest-dsvm-full http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-full/55ff837/ : SUCCESS in 52m 29s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-postgres-full/5074401/ : SUCCESS in 47m 00s\n- check-tempest-dsvm-neutron http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-neutron/a165389/ : SUCCESS in 43m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-neutron-full/1baad27/ : SUCCESS in 59m 13s\n- check-grenade-dsvm http://logs.openstack.org/57/187857/12/check/check-grenade-dsvm/88aab2d/ : SUCCESS in 52m 24s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/12/check/gate-tempest-dsvm-large-ops/92ecb93/ : SUCCESS in 36m 18s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/12/check/gate-tempest-dsvm-neutron-large-ops/a0730b2/ : SUCCESS in 31m 30s\n- check-devstack-dsvm-cells http://logs.openstack.org/57/187857/12/check/check-devstack-dsvm-cells/262eaa2/ : FAILURE in 22m 25s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/12/check/gate-nova-tox-functional/04f3a71/ : SUCCESS in 10m 01s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/12/check/check-grenade-dsvm-partial-ncpu/b786fe9/ : SUCCESS in 44m 34s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-ironic-pxe_ssh/5ecfdfb/ : SUCCESS in 44m 16s\n- check-tempest-dsvm-docker http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-docker/80e8b78/ : SUCCESS in 37m 25s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-nova-v21-full/aee88a9/ : SUCCESS in 41m 08s\n- check-tempest-dsvm-cells http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-cells/81685c1/ : SUCCESS in 37m 06s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-full-ceph/8e65cb8/ : SUCCESS in 45m 25s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-multinode-full/66f9fb0/ : FAILURE in 23m 24s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-multinode-smoke/209d68d/ : FAILURE in 23m 16s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/12/check/check-tempest-dsvm-neutron-multinode-smoke/d44a6a7/ : FAILURE in 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2910b8054902393f067526fccc77c73b2d461f78","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-09 20:47:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/12/check/dsvm-tempest-xen/a7ecde6 : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":12},{"id":"d22f7d0c8290e2b6f30617164e0a43e16d1691b7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-09 22:35:21.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/12/ : FAILURE in 2h 24m 40s","accounts_in_message":[],"_revision_number":12},{"id":"aef11c07fc35ec1c9ad7fd96e46002644e60a9f5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-10 02:39:56.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/12/Nova-ML2-Sriov : SUCCESS in 3m 07s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"0f9b39e94dbd7dae245a68e45004ca9aa70ff4de","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-10 15:37:59.000000000","message":"Patch Set 12:\n\nWe use Gerrit internally here, and if there\u0027s one thing it\u0027s shown itself to be astoundingly bad at, it\u0027s managing dependant patch sets when both are in flux. Even moreso when different authors are managing different parts of the patch train.\n\nI\u0027m OK with waiting for the other work to merge and re-working mine to account for the changes afterward, or merging mine and having the opposite happen, but attempting to build something on top of an unstable foundation is inviting spontaneous collapse.","accounts_in_message":[],"_revision_number":12},{"id":"3727be48447a973d444c4ac4f59febb842253fa7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-15 05:45:22.000000000","message":"Patch Set 12: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/12 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"fe89a50d120462d715d956fce3de0a2186141ad8","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-21 14:47:12.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"c60909a5e1cf67242d659bafa2d0f05933395a98","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-21 15:49:32.000000000","message":"Patch Set 13: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/13/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/152dc057abde40cf870cd924dcf59560/index.html : SUCCESS in 13m 35s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/13/check/real-db-upgrade_nova_mysql_user_001:th-mysql/074c12be7bcc4e379717709a30d582f9/index.html : SUCCESS in 15m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/13/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c20fe3a32e034beda820eb3826079456/index.html : SUCCESS in 14m 24s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/13/check/real-db-upgrade_nova_percona_user_001:th-percona/cab9ed1885a14a9380055a13ca96f557/index.html : SUCCESS in 16m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/13/check/real-db-upgrade_nova_mysql_user_002:th-mysql/14004baf7440482bb3d11c94ff50e63d/index.html : SUCCESS in 13m 46s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/13/check/real-db-upgrade_nova_percona_user_002:th-percona/6381eaf26d234e358aa20ea8803e9b9f/index.html : SUCCESS in 14m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":13},{"id":"344b6a2d4c58184c2dde83593738075afb70eab1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-21 16:06:20.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/13/check-ibm-tempest-dsvm-full/aafb368 : SUCCESS in 1h 17m 18s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/13/check-ibm-tempest-dsvm-postgres-full/6d16a1d : SUCCESS in 1h 09m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fa7a3c2122973fe2eb341490d3be679acdbe9b5c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-21 16:09:09.000000000","message":"Patch Set 13: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/13/5105/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":13},{"id":"af4db39a5feaa8fec51044c78e8c9c8a34e99589","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-21 16:20:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/13/check/dsvm-tempest-xen/1095c4a : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":13},{"id":"d13aee5e990fb043beabe57d1e2d97c1abc39463","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-21 17:17:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/13/check/gate-nova-pep8/e895cb2/ : SUCCESS in 7m 58s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/13/check/gate-nova-docs/48484de//doc/build/html/ : SUCCESS in 5m 53s\n- gate-nova-python27 http://logs.openstack.org/57/187857/13/check/gate-nova-python27/f735865/ : SUCCESS in 10m 41s\n- gate-nova-python34 http://logs.openstack.org/57/187857/13/check/gate-nova-python34/639979d/ : SUCCESS in 10m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-full/0cf8f4a/ : SUCCESS in 53m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-postgres-full/7f37e55/ : SUCCESS in 50m 40s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-neutron-full/a42cf80/ : SUCCESS in 59m 47s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/13/check/gate-grenade-dsvm/4ebda66/ : FAILURE in 30m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-large-ops/be9b075/ : SUCCESS in 37m 45s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-neutron-large-ops/ab85d4a/ : SUCCESS in 32m 06s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/13/check/gate-devstack-dsvm-cells/1719bcd/ : FAILURE in 28m 55s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/13/check/gate-nova-tox-functional/582cd34/ : SUCCESS in 9m 20s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/13/check/gate-grenade-dsvm-partial-ncpu/a4bd640/ : FAILURE in 27m 43s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-ironic-pxe_ssh/07a4875/ : SUCCESS in 47m 29s\n- gate-tempest-dsvm-docker http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-docker/97eee5f/ : FAILURE in 23m 21s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-nova-v21-full/c641c02/ : SUCCESS in 53m 03s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-cells/66cb8d3/ : SUCCESS in 49m 26s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-full-ceph/1e08503/ : FAILURE in 55m 05s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-multinode-full/bf92cd8/ : FAILURE in 59m 45s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-multinode-smoke/da9716c/ : SUCCESS in 46m 46s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/13/check/gate-tempest-dsvm-neutron-multinode-smoke/eb932f2/ : SUCCESS in 48m 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"70b5ec41be9231eb919f82f30c3cb43f70e29bbe","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-21 18:15:59.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/187857/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"028de4cf14b39e2694de43251c472297415ea011","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-21 19:27:25.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/13/Nova-ML2-Sriov : SUCCESS in 50m 58s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"11670c2c392fe825263e05c69b3137d7e7eb6ef0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-22 02:46:50.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/13/ : FAILURE in 2h 35m 23s","accounts_in_message":[],"_revision_number":13},{"id":"dbe36b5344e67bd801dd421d9b774c697bb30299","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-22 03:25:14.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"2fa7e50319097691b7a52c5fc724c4676a064794","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-22 04:03:59.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/187857/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"f7c9ffd27c4b5f90d7f68833320a4a59cd7b8d0a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-22 04:15:44.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/14/check-ibm-tempest-dsvm-full/e8a9d21 : SUCCESS in 49m 58s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/14/check-ibm-tempest-dsvm-postgres-full/b9dbc1d : SUCCESS in 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"da035b452cc24ca090ef72244f4dcb75dc2fd65d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-22 04:30:03.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/75e0ba147eca4dbd92bda1e20e1ce558/index.html : SUCCESS in 13m 30s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a7d02eda7d2446bf9b2e3e5bbe4238a3/index.html : SUCCESS in 15m 29s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b38a3b6f1b694bb8b9784e8316326ca8/index.html : SUCCESS in 14m 26s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/14/check/real-db-upgrade_nova_percona_user_001:th-percona/a97d0bad35114651bd3c8efbbc0d3967/index.html : SUCCESS in 15m 01s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/14/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2bc216c45fb34d93a667e4d916fdd6c0/index.html : SUCCESS in 12m 06s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/14/check/real-db-upgrade_nova_percona_user_002:th-percona/f787ce0bb57c40ac8dfdc33e3b4b9a88/index.html : SUCCESS in 13m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"eebd5d027a50aa314d3af41874e20926e465809b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-22 04:31:18.000000000","message":"Patch Set 14: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/14/5200/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":14},{"id":"7213140db7c1e858e1a3c20182f82967f81a036a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-22 04:35:54.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/14/check/dsvm-tempest-xen/f0bd558 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":14},{"id":"31d2b99a62a8317010baa7402694ad891a820ed0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-22 05:31:46.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/14/check/gate-nova-pep8/08e933d/ : SUCCESS in 8m 40s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/14/check/gate-nova-docs/c2149ab//doc/build/html/ : SUCCESS in 8m 34s\n- gate-nova-python27 http://logs.openstack.org/57/187857/14/check/gate-nova-python27/5fdc536/ : SUCCESS in 19m 44s\n- gate-nova-python34 http://logs.openstack.org/57/187857/14/check/gate-nova-python34/e003a48/ : SUCCESS in 13m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-full/4d277c7/ : SUCCESS in 54m 07s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-postgres-full/b2a5123/ : SUCCESS in 48m 53s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-neutron-full/a064cde/ : SUCCESS in 1h 06m 56s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/14/check/gate-grenade-dsvm/c97fcf6/ : SUCCESS in 45m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-large-ops/00aa7ed/ : SUCCESS in 35m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-neutron-large-ops/5ff873d/ : SUCCESS in 37m 20s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/14/check/gate-devstack-dsvm-cells/3b74481/ : FAILURE in 28m 37s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/14/check/gate-nova-tox-functional/486c489/ : SUCCESS in 13m 33s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/14/check/gate-grenade-dsvm-partial-ncpu/d6b6315/ : SUCCESS in 49m 15s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-ironic-pxe_ssh/5e3f233/ : FAILURE in 1h 56m 09s\n- gate-tempest-dsvm-docker http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-docker/1baaeb3/ : FAILURE in 21m 34s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-nova-v21-full/439d741/ : SUCCESS in 52m 27s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-cells/f93b971/ : SUCCESS in 45m 08s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-full-ceph/05ec907/ : SUCCESS in 50m 10s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-multinode-full/517a04f/ : FAILURE in 1h 07m 07s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-multinode-smoke/d5a3d67/ : SUCCESS in 43m 16s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-neutron-multinode-smoke/907e253/ : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"dcda6570e9ad74c9b3237ecd340f1eeb66748a93","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-22 05:32:40.000000000","message":"Patch Set 14:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- gate-tempest-dsvm-ironic-pxe_ssh: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":14},{"id":"cc1a03db174941cc79f14920b931fb3bf7c51b6d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-22 10:53:47.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/14/check/gate-nova-pep8/f38df79/ : NOT_REGISTERED\n- gate-nova-docs http://logs.openstack.org/57/187857/14/check/gate-nova-docs/0498282/ : NOT_REGISTERED\n- gate-nova-python27 http://logs.openstack.org/57/187857/14/check/gate-nova-python27/0028029/ : NOT_REGISTERED\n- gate-nova-python34 http://logs.openstack.org/57/187857/14/check/gate-nova-python34/771fd4e/ : NOT_REGISTERED\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-full/77986fb/ : SUCCESS in 43m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-postgres-full/197b4a8/ : SUCCESS in 48m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-neutron-full/44f98d5/ : SUCCESS in 1h 04m 43s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/14/check/gate-grenade-dsvm/dc8bc8c/ : SUCCESS in 45m 01s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-large-ops/eb0051c/ : SUCCESS in 34m 35s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-neutron-large-ops/6075828/ : SUCCESS in 25m 17s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/14/check/gate-devstack-dsvm-cells/d208914/ : FAILURE in 22m 40s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/14/check/gate-nova-tox-functional/2448577/ : NOT_REGISTERED\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/14/check/gate-grenade-dsvm-partial-ncpu/cf3bcea/ : SUCCESS in 45m 13s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-ironic-pxe_ssh/746a0d6/ : SUCCESS in 36m 52s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-nova-v21-full/0c62bfb/ : SUCCESS in 44m 41s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-cells/bcc6259/ : SUCCESS in 37m 20s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-full-ceph/e220031/ : SUCCESS in 51m 52s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-multinode-full/a598a4d/ : NOT_REGISTERED (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-multinode-smoke/19f5921/ : NOT_REGISTERED (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/14/check/gate-tempest-dsvm-neutron-multinode-smoke/72d6d3a/ : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e88114462d3a93961cff8f1de4169f4d2b58d33f","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-22 10:54:01.000000000","message":"Patch Set 14:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":14},{"id":"a890158c3179f4fb35909f76c4c1957e9fc7b1cc","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-22 13:04:11.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"0c5c2bd8e8a59d139db4a445a4f01369c692b949","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-22 13:21:25.000000000","message":"Patch Set 15: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/15/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3cadfa02af414c7ab947b7c61e54f90d/index.html : SUCCESS in 12m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/15/check/real-db-upgrade_nova_mysql_user_001:th-mysql/250a95596f2d42b1b9f8342b02d70918/index.html : SUCCESS in 13m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/15/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9b05f4e6d58e4595a171ea94d9b073a2/index.html : SUCCESS in 11m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/15/check/real-db-upgrade_nova_percona_user_001:th-percona/772580ff7ebd40a286ffd720487fd49e/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/15/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1026519bc16d48299da077d3df01a202/index.html : SUCCESS in 13m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/15/check/real-db-upgrade_nova_percona_user_002:th-percona/da57662666ee48a9889c6ce22f2ef1e6/index.html : SUCCESS in 12m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":15},{"id":"f55fc804200ad7a0230b547028897611fb452eef","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-22 13:27:28.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/187857/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"ca10ba170e6a0bfeae19f1677eae8f0b77359bae","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-22 13:59:40.000000000","message":"Patch Set 15: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/15/5255/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":15},{"id":"ab0971f701d5ba5e7c45b1e86b3924ac77a251c4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-22 14:12:07.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/15/check-ibm-tempest-dsvm-full/95f0b48 : SUCCESS in 48m 26s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/15/check-ibm-tempest-dsvm-postgres-full/9d988d5 : SUCCESS in 1h 07m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3ee90ddb8b3b2f89c722cda94f6253053f0ac058","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-22 14:21:26.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/15/check/dsvm-tempest-xen/7e61381 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":15},{"id":"7824538a3218f6ca29c496928e930f2e5f48a748","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-22 14:36:49.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/15/check/gate-nova-pep8/d4b9536/ : SUCCESS in 9m 18s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/15/check/gate-nova-docs/8bf8ee5//doc/build/html/ : SUCCESS in 6m 43s\n- gate-nova-python27 http://logs.openstack.org/57/187857/15/check/gate-nova-python27/35500ce/ : SUCCESS in 11m 31s\n- gate-nova-python34 http://logs.openstack.org/57/187857/15/check/gate-nova-python34/7a3302a/ : SUCCESS in 13m 36s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-full/68391aa/ : SUCCESS in 58m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-postgres-full/aee089f/ : SUCCESS in 41m 25s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-neutron-full/46bfc4f/ : SUCCESS in 1h 07m 29s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/15/check/gate-grenade-dsvm/a236450/ : SUCCESS in 46m 32s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-large-ops/7ad7eef/ : SUCCESS in 35m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-neutron-large-ops/864dee2/ : SUCCESS in 30m 40s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/15/check/gate-devstack-dsvm-cells/03bd64f/ : FAILURE in 34m 47s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/15/check/gate-nova-tox-functional/feec21d/ : SUCCESS in 13m 17s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/15/check/gate-grenade-dsvm-partial-ncpu/4ff086c/ : SUCCESS in 46m 05s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-ironic-pxe_ssh/7a23165/ : SUCCESS in 40m 45s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-nova-v21-full/cf7ec6e/ : SUCCESS in 49m 41s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-cells/6bf1f10/ : SUCCESS in 51m 09s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-full-ceph/e54b373/ : SUCCESS in 56m 50s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-multinode-full/4bff8e9/ : SUCCESS in 52m 00s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-multinode-smoke/2a080ad/ : SUCCESS in 35m 23s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-neutron-multinode-smoke/fbadcf7/ : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c138149a6e810b776ff1ad15b931bbf9e7129b65","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-22 14:37:01.000000000","message":"Patch Set 15:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":15},{"id":"ea1f9ce8f5998c4acd33005a8f464d9180cb249c","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-22 14:39:39.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"4b4219706302a02d791659802593d613c3228a58","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-22 14:50:56.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/15/ : FAILURE in 7m 19s","accounts_in_message":[],"_revision_number":15},{"id":"ad8199c17177c343b0baaef1199300a36e673a74","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-22 18:53:26.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"a061f6ca126cf6b40502ae8fb5a8541409420682","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-22 19:53:43.000000000","message":"Patch Set 16: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/16/5320/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":16},{"id":"41b1ddca101ac5a38f6e4f4f16670245e1bdb1df","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-22 20:03:29.000000000","message":"Patch Set 16:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/16/check-ibm-tempest-dsvm-full/55815fe : SUCCESS in 50m 21s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/16/check-ibm-tempest-dsvm-postgres-full/c581ba7 : SUCCESS in 1h 09m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e3126f2cb79ba2c37de558332f598a2338e5a1c1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-22 20:35:00.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/16/ : FAILURE in 1h 40m 43s","accounts_in_message":[],"_revision_number":16},{"id":"e583d534d3ae366e6256d8141091582f2b6dcfbb","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-22 20:59:20.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/16/check/dsvm-tempest-xen/0043267 : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":16},{"id":"3b0e0a50ec6f75464dbdb0cc470d89133a9daf26","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-22 21:43:57.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/187857/16 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"8fe1dd2292b7cd30e4e334e8ce7ca631cb359a9b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-22 22:26:40.000000000","message":"Patch Set 16: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/16/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f66333d362864ce6acd4f06180c3afda/index.html : SUCCESS in 13m 53s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/16/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4b78f59d18ef4fe9bef27bb414f59554/index.html : SUCCESS in 14m 50s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/16/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6a4381720b794da6be29459290219b26/index.html : SUCCESS in 15m 06s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/16/check/real-db-upgrade_nova_percona_user_001:th-percona/7acf489eb09e4a458b58b17f569fa46c/index.html : SUCCESS in 15m 51s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/16/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0a834237fba142098eca308eac6c825a/index.html : SUCCESS in 14m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/16/check/real-db-upgrade_nova_percona_user_002:th-percona/fe74becb1b7e480a91be2f1091f4f540/index.html : SUCCESS in 14m 16s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":16},{"id":"5141f104c66731d936585d8c11dd3d1d9b65f345","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 02:28:39.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/16/check/gate-nova-pep8/fcfffef/ : SUCCESS in 8m 46s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/16/check/gate-nova-docs/53aecef//doc/build/html/ : SUCCESS in 6m 22s\n- gate-nova-python27 http://logs.openstack.org/57/187857/16/check/gate-nova-python27/3b83bb4/ : SUCCESS in 13m 17s\n- gate-nova-python34 http://logs.openstack.org/57/187857/16/check/gate-nova-python34/1a4de15/ : SUCCESS in 9m 12s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-full/f72d76d/ : SUCCESS in 58m 07s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-postgres-full/82e846b/ : SUCCESS in 37m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-neutron-full/fa65e47/ : SUCCESS in 47m 36s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/16/check/gate-grenade-dsvm/a1d1305/ : SUCCESS in 45m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-large-ops/192e9c9/ : SUCCESS in 29m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-neutron-large-ops/e2877af/ : SUCCESS in 28m 30s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/16/check/gate-devstack-dsvm-cells/8cdc4dd/ : FAILURE in 22m 24s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/16/check/gate-nova-tox-functional/345f745/ : SUCCESS in 12m 32s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/16/check/gate-grenade-dsvm-partial-ncpu/5e45c79/ : SUCCESS in 51m 46s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-ironic-pxe_ssh/dd614fb/ : SUCCESS in 41m 04s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-nova-v21-full/1ffc521/ : SUCCESS in 52m 56s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-cells/4c60a77/ : SUCCESS in 55m 58s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-full-ceph/4221156/ : SUCCESS in 51m 38s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-multinode-full/a4152a1/ : SUCCESS in 49m 08s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-multinode-smoke/7570046/ : SUCCESS in 34m 48s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/16/check/gate-tempest-dsvm-neutron-multinode-smoke/0803b16/ : SUCCESS in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"93907633ad4a216e6ec817f9baab1349bfcb0bf8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 03:16:45.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/15/check/gate-nova-pep8/8bceef2/ : SUCCESS in 6m 50s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/15/check/gate-nova-docs/6598983//doc/build/html/ : SUCCESS in 8m 45s\n- gate-nova-python27 http://logs.openstack.org/57/187857/15/check/gate-nova-python27/8e162cf/ : SUCCESS in 10m 19s\n- gate-nova-python34 http://logs.openstack.org/57/187857/15/check/gate-nova-python34/ae14fa6/ : SUCCESS in 10m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-full/e8f4aef/ : SUCCESS in 41m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-postgres-full/c14aa17/ : FAILURE in 49m 18s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-neutron-full/6968fdf/ : SUCCESS in 57m 05s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/15/check/gate-grenade-dsvm/b700b09/ : SUCCESS in 46m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-large-ops/f7556b0/ : SUCCESS in 32m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-neutron-large-ops/253f900/ : SUCCESS in 30m 51s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/15/check/gate-devstack-dsvm-cells/5411372/ : FAILURE in 28m 03s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/15/check/gate-nova-tox-functional/2eed0b3/ : SUCCESS in 10m 47s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/15/check/gate-grenade-dsvm-partial-ncpu/0cba956/ : SUCCESS in 58m 33s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-ironic-pxe_ssh/83b2cce/ : SUCCESS in 41m 37s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-nova-v21-full/fc32306/ : SUCCESS in 50m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-cells/9eb100e/ : SUCCESS in 50m 24s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-full-ceph/bd544b2/ : SUCCESS in 50m 53s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-multinode-full/1ac44cb/ : FAILURE in 1h 13m 11s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-multinode-smoke/56968c5/ : SUCCESS in 43m 08s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/15/check/gate-tempest-dsvm-neutron-multinode-smoke/462f71b/ : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"36367eb163f5ca8d8cb8273b972990a4f9df267e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-23 07:53:10.000000000","message":"Patch Set 16:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":16},{"id":"87517e52e9f622646278c52e8b6bd50224c4ac43","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-23 14:17:33.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"ce07b2fb75e9b742ee6e7407635b43220ff4b410","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-23 14:33:17.000000000","message":"Patch Set 17: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/17/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/14d99571b47940cdb610c5405fd07179/index.html : SUCCESS in 11m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/17/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e1557b3d9f0742a1ac92cd915218dcc6/index.html : SUCCESS in 14m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/17/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1a15b345ec56477f9f1cefcaf8562085/index.html : SUCCESS in 11m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/17/check/real-db-upgrade_nova_percona_user_001:th-percona/510ad0519ccd40569103126340bf8f67/index.html : SUCCESS in 12m 06s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/17/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7817ba1adc4e49f1958c8339cd5ed144/index.html : SUCCESS in 12m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/17/check/real-db-upgrade_nova_percona_user_002:th-percona/81acee5ec5e94528971210ccfaae6caa/index.html : SUCCESS in 14m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":17},{"id":"b8cf6716d289cc9e2a49053b76d2ccdb06c605b0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-23 15:04:17.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/187857/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"a820b19aec566945f084891edf87a18acee6c1f7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-23 15:25:48.000000000","message":"Patch Set 17: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/17/5437/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":17},{"id":"a0093d7aff3cb0479eda9bce742dc2b704bf60c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-23 15:31:01.000000000","message":"Patch Set 17:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/17/check-ibm-tempest-dsvm-full/56102c4 : SUCCESS in 1h 12m 55s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/17/check-ibm-tempest-dsvm-postgres-full/c4c55b3 : SUCCESS in 47m 03s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7ba3864b725cdd57b4252a46f218251ae13e95b9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-23 15:38:50.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/17/check/dsvm-tempest-xen/7c53781 : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":17},{"id":"4b79e4cd14b197d6547beeedc8ba60a6f11f05dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 15:58:18.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/17/check/gate-nova-pep8/80df3ec/ : SUCCESS in 6m 46s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/17/check/gate-nova-docs/65256be//doc/build/html/ : SUCCESS in 8m 14s\n- gate-nova-python27 http://logs.openstack.org/57/187857/17/check/gate-nova-python27/3df45ce/ : SUCCESS in 10m 28s\n- gate-nova-python34 http://logs.openstack.org/57/187857/17/check/gate-nova-python34/7694fa5/ : SUCCESS in 9m 56s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-full/01c916b/ : SUCCESS in 47m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-postgres-full/b95e027/ : SUCCESS in 48m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-neutron-full/9bbff8e/ : SUCCESS in 1h 04m 18s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/17/check/gate-grenade-dsvm/310301c/ : SUCCESS in 1h 04m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-large-ops/1aac8a1/ : SUCCESS in 34m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-neutron-large-ops/81973f9/ : SUCCESS in 32m 55s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/17/check/gate-devstack-dsvm-cells/aead6c2/ : FAILURE in 24m 52s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/17/check/gate-nova-tox-functional/fb33a31/ : SUCCESS in 9m 18s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/17/check/gate-grenade-dsvm-partial-ncpu/a4ed153/ : SUCCESS in 49m 37s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-ironic-pxe_ssh/0303ec6/ : SUCCESS in 41m 50s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-nova-v21-full/b06c7ca/ : SUCCESS in 51m 49s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-cells/08c7180/ : SUCCESS in 53m 08s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-full-ceph/5403acf/ : SUCCESS in 51m 36s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-multinode-full/ffdf9eb/ : SUCCESS in 50m 44s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-multinode-smoke/81df858/ : SUCCESS in 49m 10s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/17/check/gate-tempest-dsvm-neutron-multinode-smoke/3eac7f1/ : FAILURE in 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d67c4c1914a4571313ae70f9cfab1c12de19e3eb","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-24 02:09:24.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"80c5e047ca8c09fdabfdb7820ced76f663eaaf99","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-24 03:06:03.000000000","message":"Patch Set 18: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/18/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/649b7177bc9848b680ec50476118b0b8/index.html : SUCCESS in 14m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/18/check/real-db-upgrade_nova_mysql_user_001:th-mysql/115aac063505473bb972630966df29ac/index.html : SUCCESS in 16m 21s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/18/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/99acd265622549009724c67b2c416386/index.html : SUCCESS in 11m 54s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/18/check/real-db-upgrade_nova_percona_user_001:th-percona/528826441a39470ba1756cfb65f22b14/index.html : SUCCESS in 15m 28s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/18/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0f301b93e0b34d43b7ce3cc5734e20bd/index.html : SUCCESS in 14m 46s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/18/check/real-db-upgrade_nova_percona_user_002:th-percona/94e0755bcc104db399d15b70423ab849/index.html : SUCCESS in 13m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":18},{"id":"9806f4a0abbd72f0157c094148a7f3becd3ed88b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-24 03:11:05.000000000","message":"Patch Set 18: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/18/5566/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":18},{"id":"d63d6c567ad85e81a1a3d447fd7007b3555ea5f4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-24 03:22:02.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/18/check-ibm-tempest-dsvm-full/1ad118a : SUCCESS in 1h 12m 11s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/18/check-ibm-tempest-dsvm-postgres-full/a1f9d5c : SUCCESS in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"eb3d42fc2cc5749ba343e3ba427957a52de2a006","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-24 03:39:02.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/18/check/gate-nova-pep8/4608d8d/ : SUCCESS in 9m 33s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/18/check/gate-nova-docs/8b9cc92//doc/build/html/ : SUCCESS in 9m 00s\n- gate-nova-python27 http://logs.openstack.org/57/187857/18/check/gate-nova-python27/c33f381/ : FAILURE in 13m 54s\n- gate-nova-python34 http://logs.openstack.org/57/187857/18/check/gate-nova-python34/823b3cd/ : SUCCESS in 13m 03s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-full/f885aa3/ : SUCCESS in 54m 54s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-postgres-full/760a225/ : SUCCESS in 58m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-neutron-full/531cc20/ : SUCCESS in 1h 14m 01s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/18/check/gate-grenade-dsvm/31bad6e/ : SUCCESS in 51m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-large-ops/4566f57/ : SUCCESS in 35m 44s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-neutron-large-ops/c85743a/ : SUCCESS in 32m 48s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/18/check/gate-devstack-dsvm-cells/2aef60e/ : FAILURE in 29m 01s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/18/check/gate-nova-tox-functional/5501599/ : SUCCESS in 15m 09s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/18/check/gate-grenade-dsvm-partial-ncpu/b0fd721/ : SUCCESS in 46m 54s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-ironic-pxe_ssh/5ca0264/ : SUCCESS in 44m 15s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-nova-v21-full/7a33e65/ : SUCCESS in 55m 35s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-cells/5b314d2/ : SUCCESS in 38m 49s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-full-ceph/4325952/ : SUCCESS in 1h 08m 40s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-multinode-full/22b5304/ : SUCCESS in 58m 40s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-multinode-smoke/72decb6/ : SUCCESS in 45m 15s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/18/check/gate-tempest-dsvm-neutron-multinode-smoke/7739652/ : SUCCESS in 50m 59s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"9299b19dd844513b7f486bcc00d3dcd51e5b15a7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-24 03:49:35.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/187857/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"cad4c65df240efb20eff2f428dc69906ecdbde30","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-24 03:58:37.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/18/ : FAILURE in 1h 48m 15s","accounts_in_message":[],"_revision_number":18},{"id":"77a4d143a35c079c48309161d56f6aa34a3cf958","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-24 04:11:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/18/check/dsvm-tempest-xen/dd9e2a4 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":18},{"id":"a73a3efaab07d8f259e79cc5c1225b0c023188b7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-24 04:33:58.000000000","message":"Patch Set 18:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- gate-nova-python27: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":18},{"id":"d96ac30b793c7bd10d71b2f1f9d60631d950d03c","author":{"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},"date":"2015-07-24 08:06:19.000000000","message":"Patch Set 18: Code-Review-1","accounts_in_message":[],"_revision_number":18},{"id":"4420709235e9756a958cb331e297be953032e552","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-24 16:50:06.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"aa193fe0d1df7faafbac8c76bce3ba79989264a6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-24 18:31:14.000000000","message":"Patch Set 19:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/19/check-ibm-tempest-dsvm-full/c536939 : SUCCESS in 1h 12m 52s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/19/check-ibm-tempest-dsvm-postgres-full/4ccf697 : SUCCESS in 1h 08m 29s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"e1c2010cd8b339e9cfa99c23007ae3bafb40ce0a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-24 19:16:51.000000000","message":"Patch Set 19: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/19/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a57d6656e3a1482d9d1d751057257cc8/index.html : SUCCESS in 12m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/19/check/real-db-upgrade_nova_mysql_user_001:th-mysql/74e79dcafe654ed69c3ee020f85249f1/index.html : SUCCESS in 12m 42s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/19/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/768821741dda45a6b6383f696af74855/index.html : SUCCESS in 15m 45s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/19/check/real-db-upgrade_nova_percona_user_001:th-percona/abbb03d27e5c4308b079af8b9a86997e/index.html : SUCCESS in 17m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/19/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e7ec5e5e37c34dbda965f2bcee25ad1b/index.html : SUCCESS in 14m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/19/check/real-db-upgrade_nova_percona_user_002:th-percona/f3b7166a9dde412f911d75730db92b58/index.html : SUCCESS in 14m 35s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":19},{"id":"22ba66a6aabf6bee6205f614ee6da069b6b9a3ef","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-24 19:26:25.000000000","message":"Patch Set 19: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/19/5685/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":19},{"id":"aaaed0373d6e14dcfcf956c5629711839690ff99","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-24 19:33:32.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/19/check/dsvm-tempest-xen/c5d8ed2 : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":19},{"id":"8eb85b640b57932502f2adb9f9ecc23079899b8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-24 19:43:33.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/19/check/gate-nova-pep8/02893aa/ : SUCCESS in 7m 01s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/19/check/gate-nova-docs/6d18ca6//doc/build/html/ : SUCCESS in 6m 55s\n- gate-nova-python27 http://logs.openstack.org/57/187857/19/check/gate-nova-python27/89afa62/ : SUCCESS in 14m 14s\n- gate-nova-python34 http://logs.openstack.org/57/187857/19/check/gate-nova-python34/4e472ec/ : SUCCESS in 9m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-full/19823b5/ : SUCCESS in 59m 18s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-postgres-full/ab33e95/ : SUCCESS in 57m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-neutron-full/b25b7ff/ : SUCCESS in 1h 07m 40s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/19/check/gate-grenade-dsvm/f8ccaab/ : SUCCESS in 50m 28s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-large-ops/e2ca094/ : SUCCESS in 26m 25s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-neutron-large-ops/481f55c/ : SUCCESS in 26m 53s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/19/check/gate-devstack-dsvm-cells/8966479/ : FAILURE in 27m 43s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/19/check/gate-nova-tox-functional/752698d/ : SUCCESS in 11m 56s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/19/check/gate-grenade-dsvm-partial-ncpu/692ce86/ : SUCCESS in 48m 29s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-ironic-pxe_ssh/c22eed7/ : SUCCESS in 38m 02s\n- gate-tempest-dsvm-ironic-pxe_ssh-full-nv http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-ironic-pxe_ssh-full-nv/4892f28/ : FAILURE in 1h 47m 19s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-nova-v21-full/6b7be7f/ : SUCCESS in 42m 08s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-cells/db984c0/ : SUCCESS in 39m 16s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-full-ceph/9d54565/ : SUCCESS in 58m 28s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-multinode-full/fc03132/ : FAILURE in 51m 35s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-multinode-smoke/1f144f7/ : SUCCESS in 34m 42s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/19/check/gate-tempest-dsvm-neutron-multinode-smoke/1eec4e3/ : SUCCESS in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"29011381b694ab26907135548c0d422d52f8b6c4","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-24 19:43:45.000000000","message":"Patch Set 19:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":19},{"id":"086db51b2e9b83681a811f7d74eaf029a4ae5ec5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-24 19:59:39.000000000","message":"Patch Set 19:\n\n* pci-test http://52.27.155.124/187857/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"636aa77798c43cc8c1b5c4362bf650ff659631c6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-24 23:24:02.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/19/ : FAILURE in 1h 49m 51s","accounts_in_message":[],"_revision_number":19},{"id":"4d7a1a5390581ed34b36799ee449a269879ab151","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-27 01:28:03.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"b39e075e0525484bc72ec668fb8fe05244672316","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-27 01:28:19.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"23fc77486722ea6ccac7718cf579aa0526d2ff51","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-27 01:45:04.000000000","message":"Patch Set 20:\n\n* pci-test http://52.27.155.124/187857/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"7366721a96646062389d4f7ff90e593d588fc177","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-27 01:45:10.000000000","message":"Patch Set 20: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/20/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b1b217df2d36470999f91f7501c7763f/index.html : SUCCESS in 13m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/20/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d05d4ab5f4c34a12aa366390c379062c/index.html : SUCCESS in 14m 20s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/20/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/aadc274da4f44234a03be0e060e36676/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/20/check/real-db-upgrade_nova_percona_user_001:th-percona/3f83888681c544f0856889d738c78ba9/index.html : SUCCESS in 16m 20s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/20/check/real-db-upgrade_nova_mysql_user_002:th-mysql/dda6f74344c446b194009a84aee16670/index.html : SUCCESS in 11m 43s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/20/check/real-db-upgrade_nova_percona_user_002:th-percona/b56d1007945b4ba1b2336340b7c4ecee/index.html : SUCCESS in 13m 22s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":20},{"id":"14bb54fc219fbb33b2116d618a9cfa26683795c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-27 02:16:15.000000000","message":"Patch Set 20:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/20/check-ibm-tempest-dsvm-full/d9e430f : SUCCESS in 47m 45s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/20/check-ibm-tempest-dsvm-postgres-full/9ff8234 : SUCCESS in 46m 08s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"71fefa5541c359d16887be39c3ae42c98f6ebf11","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-27 02:29:46.000000000","message":"Patch Set 20: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/20/5804/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":20},{"id":"8f25f92530e88ed88de093416fe18453e111977a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-27 03:05:33.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/20/ : FAILURE in 1h 36m 36s","accounts_in_message":[],"_revision_number":20},{"id":"c3292da7b70f57de9f50a968b8c46864f8dda0cc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-27 03:29:30.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/20/check/gate-nova-pep8/dfd82a6/ : SUCCESS in 7m 06s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/20/check/gate-nova-docs/bf96087//doc/build/html/ : SUCCESS in 7m 00s\n- gate-nova-python27 http://logs.openstack.org/57/187857/20/check/gate-nova-python27/1bf9bb4/ : SUCCESS in 12m 17s\n- gate-nova-python34 http://logs.openstack.org/57/187857/20/check/gate-nova-python34/8994521/ : SUCCESS in 10m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-full/b6c4899/ : SUCCESS in 52m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-postgres-full/57427dc/ : SUCCESS in 52m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-neutron-full/59356c0/ : SUCCESS in 1h 02m 35s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/20/check/gate-grenade-dsvm/f0c1fbb/ : SUCCESS in 42m 10s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-large-ops/67f3076/ : SUCCESS in 33m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-neutron-large-ops/272ce9e/ : SUCCESS in 25m 24s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/20/check/gate-devstack-dsvm-cells/3a6592b/ : FAILURE in 30m 46s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/20/check/gate-nova-tox-functional/fcf0d8a/ : SUCCESS in 11m 27s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/20/check/gate-grenade-dsvm-partial-ncpu/36d7757/ : SUCCESS in 46m 15s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-ironic-pxe_ssh/7109c23/ : SUCCESS in 48m 48s\n- gate-tempest-dsvm-ironic-pxe_ssh-full-nv http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-ironic-pxe_ssh-full-nv/1922396/ : FAILURE in 1h 57m 06s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-nova-v21-full/cdcbc53/ : SUCCESS in 52m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-cells/75a2a0a/ : SUCCESS in 47m 01s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-full-ceph/c341b39/ : SUCCESS in 55m 36s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-multinode-full/21013e5/ : SUCCESS in 54m 15s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-multinode-smoke/2b28e63/ : SUCCESS in 34m 24s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-neutron-multinode-smoke/bd8f1c5/ : SUCCESS in 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"b4c1386b44cd32b3953d577b175ce4461e81b75a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-27 03:29:43.000000000","message":"Patch Set 20:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":20},{"id":"c11c58b40bfb3e5702cc51e8f405a2b561ec57df","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-27 18:47:42.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"3f109970a5c691b494f7aa36feebd116a6c9e807","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-27 23:51:46.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/20/check/gate-nova-pep8/968c9b4/ : SUCCESS in 9m 11s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/20/check/gate-nova-docs/ee27b5a//doc/build/html/ : SUCCESS in 8m 45s\n- gate-nova-python27 http://logs.openstack.org/57/187857/20/check/gate-nova-python27/730d462/ : SUCCESS in 13m 48s\n- gate-nova-python34 http://logs.openstack.org/57/187857/20/check/gate-nova-python34/da64839/ : SUCCESS in 13m 10s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-full/06858bb/ : SUCCESS in 59m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-postgres-full/3808d4a/ : SUCCESS in 41m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-neutron-full/23a5d31/ : SUCCESS in 1h 05m 40s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/20/check/gate-grenade-dsvm/d0fbf63/ : SUCCESS in 48m 04s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-large-ops/ede70d8/ : SUCCESS in 34m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-neutron-large-ops/0135247/ : SUCCESS in 32m 47s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/20/check/gate-devstack-dsvm-cells/57183e8/ : FAILURE in 30m 16s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/20/check/gate-nova-tox-functional/a1de6b4/ : SUCCESS in 13m 55s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/20/check/gate-grenade-dsvm-partial-ncpu/68ff7e8/ : SUCCESS in 50m 29s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-ironic-pxe_ssh/8187247/ : SUCCESS in 42m 03s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-nova-v21-full/bc5a1a3/ : SUCCESS in 1h 02m 59s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-cells/39e365f/ : SUCCESS in 45m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-full-ceph/1af7347/ : SUCCESS in 1h 42m 13s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-multinode-full/cd92784/ : SUCCESS in 1h 05m 52s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-multinode-smoke/f72588e/ : SUCCESS in 45m 58s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/20/check/gate-tempest-dsvm-neutron-multinode-smoke/43a37bf/ : SUCCESS in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"137f990814e0adac0de61700274917e6cc1ad16c","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-27 23:51:57.000000000","message":"Patch Set 20:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":20},{"id":"e8f6beabeb93436b0db288101879f22ffb9fd0dd","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-28 00:12:21.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"61d23db3dde1fc60b34c85c281840acf94b847c1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-28 00:19:42.000000000","message":"Patch Set 21:\n\n* pci-test http://52.27.155.124/187857/21 : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"a9987ec4ec4bf338fffc589e00bb0328b4f82a66","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-28 01:01:43.000000000","message":"Patch Set 21:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/21/check-ibm-tempest-dsvm-full/3cb7028 : SUCCESS in 48m 53s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/21/check-ibm-tempest-dsvm-postgres-full/d7a889b : SUCCESS in 45m 11s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"9444964fcb5eca347663d3541e15810cfc71da62","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-28 02:17:40.000000000","message":"Patch Set 21:\n\n* pci-test http://52.27.155.124/187857/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"8c3a86bb90172874fb963da55d80aae4ea40e997","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-28 04:22:07.000000000","message":"Patch Set 21: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/21/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0ba0bc15d67a43ad9fbcea1a7542d98d/index.html : SUCCESS in 12m 29s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/21/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6105361d0a72417db0e9b7848b00dba5/index.html : SUCCESS in 15m 15s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/21/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1a691649a93c4f039c21d057454b3956/index.html : SUCCESS in 14m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/21/check/real-db-upgrade_nova_percona_user_001:th-percona/c514b9888e9b415fb98c64f05003e346/index.html : SUCCESS in 15m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/21/check/real-db-upgrade_nova_mysql_user_002:th-mysql/9e86c5f45b29467fbc9b805a629118ee/index.html : SUCCESS in 13m 32s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/21/check/real-db-upgrade_nova_percona_user_002:th-percona/4eb400ffcf0b4c899befea8390a84f76/index.html : SUCCESS in 12m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":21},{"id":"e8a59ac2e7916c2ed27b6594b520e98cd3eb135d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-28 05:45:09.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/21/check/dsvm-tempest-xen/bfba797 : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":21},{"id":"44087197e4b577138636baf95a25171ec234b79d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-28 12:40:10.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/21/check/gate-nova-pep8/817acff/ : SUCCESS in 8m 31s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/21/check/gate-nova-docs/7efb038//doc/build/html/ : SUCCESS in 9m 52s\n- gate-nova-python27 http://logs.openstack.org/57/187857/21/check/gate-nova-python27/2e76df6/ : SUCCESS in 14m 30s\n- gate-nova-python34 http://logs.openstack.org/57/187857/21/check/gate-nova-python34/8178c05/ : SUCCESS in 14m 34s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-full/5a9ac30/ : SUCCESS in 43m 41s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-postgres-full/40637b9/ : SUCCESS in 43m 48s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-neutron-full/2795078/ : SUCCESS in 52m 50s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/21/check/gate-grenade-dsvm/9dda0b6/ : SUCCESS in 43m 23s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-large-ops/5d4ecdb/ : SUCCESS in 26m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-neutron-large-ops/e05a684/ : SUCCESS in 27m 03s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/21/check/gate-devstack-dsvm-cells/4e0d2a8/ : SUCCESS in 24m 24s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/21/check/gate-nova-tox-functional/4926824/ : SUCCESS in 12m 49s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/21/check/gate-grenade-dsvm-partial-ncpu/cfd8bd7/ : SUCCESS in 46m 12s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-ironic-pxe_ssh/55e0017/ : SUCCESS in 38m 39s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-nova-v21-full/b0dbe0d/ : SUCCESS in 41m 35s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-cells/b2b61e0/ : SUCCESS in 39m 08s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-full-ceph/d0c10f5/ : SUCCESS in 40m 24s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-multinode-full/a622c3a/ : SUCCESS in 48m 01s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-multinode-smoke/03ad1ae/ : SUCCESS in 36m 35s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/21/check/gate-tempest-dsvm-neutron-multinode-smoke/a030139/ : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"8e9cf4f7951abed99e8b9c64d15361efd0faf8ff","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-07-28 13:46:23.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/21/check/dsvm-tempest-cinder/2206f8a/ : SUCCESS in 33m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"7da4f455d3ccf6f4aec8fcae3adfd823e9c5894e","author":{"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},"date":"2015-07-28 14:16:01.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"681dd0be2fe4b5c6d9a42d003cc9181ce1e8e6ba","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-07-28 16:06:52.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"5af7b1730c236bdc947fe2c4e5374e81e78de626","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-28 17:46:50.000000000","message":"Patch Set 22: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/22/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6a08eeeeb29e40788cd00b5e6ee67370/index.html : SUCCESS in 12m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/22/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7841f726741f4a5bb9aa3f320ae41d56/index.html : SUCCESS in 15m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/22/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5b6ccb10cda54a9291aa4a8f3607ff44/index.html : SUCCESS in 14m 11s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/22/check/real-db-upgrade_nova_percona_user_001:th-percona/a7592f08b28e469fa50bb0af18e8cef8/index.html : SUCCESS in 13m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/22/check/real-db-upgrade_nova_mysql_user_002:th-mysql/12f9a0dfb18242f0880cb9799e26c464/index.html : SUCCESS in 13m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/22/check/real-db-upgrade_nova_percona_user_002:th-percona/81043f2641a44a1dad18f16f2458278f/index.html : SUCCESS in 12m 24s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":22},{"id":"de9544d59c593aeb2f055b2f63144cb36f13ad89","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-28 20:43:16.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/22/check/dsvm-tempest-xen/16a5078 : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":22},{"id":"b109aa47bfe50df25e22d93f068a6ca7ac00106d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-28 21:03:32.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/187857/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"29294ceb031d29ca79066f38eaf07c766cee148b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-28 21:16:11.000000000","message":"Patch Set 22:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/22/check-ibm-tempest-dsvm-full/3d39334 : SUCCESS in 49m 00s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/22/check-ibm-tempest-dsvm-postgres-full/6c6dd5b : SUCCESS in 47m 48s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"c8e3b142156c960501b5b44f2548ec6cda4ec6c9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-28 21:26:39.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/22/check/gate-nova-pep8/1c4877f/ : SUCCESS in 9m 25s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/22/check/gate-nova-docs/e2d9af9//doc/build/html/ : SUCCESS in 10m 11s\n- gate-nova-python27 http://logs.openstack.org/57/187857/22/check/gate-nova-python27/9cc917c/ : SUCCESS in 13m 55s\n- gate-nova-python34 http://logs.openstack.org/57/187857/22/check/gate-nova-python34/9b8ff2c/ : SUCCESS in 13m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-full/495a2fe/ : SUCCESS in 42m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-postgres-full/5fbc9a2/ : SUCCESS in 51m 14s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-neutron-full/0091fc0/ : SUCCESS in 51m 07s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/22/check/gate-grenade-dsvm/da36dd3/ : SUCCESS in 45m 10s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-large-ops/9cd33f5/ : SUCCESS in 26m 45s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-neutron-large-ops/39beb84/ : SUCCESS in 29m 55s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/22/check/gate-devstack-dsvm-cells/84fd991/ : SUCCESS in 26m 18s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/22/check/gate-nova-tox-functional/e7aee1d/ : SUCCESS in 12m 19s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/22/check/gate-grenade-dsvm-partial-ncpu/4c36dad/ : SUCCESS in 44m 12s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-ironic-pxe_ssh/ff2235c/ : SUCCESS in 47m 03s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-nova-v21-full/da3060f/ : SUCCESS in 59m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-cells/dda4699/ : SUCCESS in 36m 42s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-full-ceph/d197a34/ : SUCCESS in 54m 23s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-multinode-full/e65f559/ : SUCCESS in 51m 55s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-multinode-smoke/6f248b5/ : SUCCESS in 42m 30s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-neutron-multinode-smoke/ccf2e34/ : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"3212bcaadf9feb9811bf1cf2851c361e23af87be","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-07-29 00:41:35.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/22/check/dsvm-tempest-cinder/eb2558a/ : FAILURE in 56m 09s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"3fbae9a8b4e2129ed71567298a5f97098e9aee95","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-29 03:09:34.000000000","message":"Patch Set 22: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/22/6160/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":22},{"id":"7680cfc93274c982ce2fa38419a0acf086fe32fa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-29 16:43:15.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/22/Nova-ML2-Sriov : SUCCESS in 1h 18m 16s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"89abc3c7c6f6f5587258d7e04c23270655ef4495","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-31 04:12:59.000000000","message":"Patch Set 22: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/22 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"5487bdf8d48ab8ad4d455dd88cee714ed1d294d2","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2015-08-08 11:45:29.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"50907e07d6a8b756beeffbfbe2ce121f5232aac5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-08 12:55:59.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/22/check/gate-nova-pep8/296e1b6/ : SUCCESS in 8m 57s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/22/check/gate-nova-docs/365a2e0//doc/build/html/ : SUCCESS in 14m 05s\n- gate-nova-python27 http://logs.openstack.org/57/187857/22/check/gate-nova-python27/dbae837/ : SUCCESS in 15m 49s\n- gate-nova-python34 http://logs.openstack.org/57/187857/22/check/gate-nova-python34/fe16d49/ : SUCCESS in 13m 19s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-full/fd70ac3/ : SUCCESS in 50m 21s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-postgres-full/ec0953c/ : SUCCESS in 48m 30s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-neutron-full/07f41f5/ : SUCCESS in 51m 48s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/22/check/gate-grenade-dsvm/c7e5458/ : SUCCESS in 46m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-large-ops/b3e622b/ : SUCCESS in 33m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-neutron-large-ops/6adf332/ : SUCCESS in 26m 57s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/22/check/gate-devstack-dsvm-cells/5ff399e/ : SUCCESS in 26m 19s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/22/check/gate-nova-tox-functional/583715a/ : SUCCESS in 12m 58s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/22/check/gate-grenade-dsvm-partial-ncpu/6123647/ : SUCCESS in 47m 29s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-ironic-pxe_ssh/72a38ec/ : SUCCESS in 44m 47s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-nova-v21-full/b881cfd/ : SUCCESS in 50m 02s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-cells/b180992/ : SUCCESS in 44m 17s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-full-ceph/34b8e75/ : SUCCESS in 43m 31s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-multinode-full/87c9c8a/ : SUCCESS in 59m 08s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/57/187857/22/check/gate-tempest-dsvm-neutron-multinode-full/14839c4/ : SUCCESS in 1h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"0879c83fc7980348587e7a4dc8ff81012dc86502","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-08-08 13:25:46.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/22/ : FAILURE in 1h 38m 24s","accounts_in_message":[],"_revision_number":22},{"id":"be6b5db28744792dbd1b79b5bd99f5603a5c7b53","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-08-08 13:32:49.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/22/check/dsvm-tempest-cinder/41f388f/ : SUCCESS in 35m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"4da7358f6b49c29b454d845c5b00099e7c8d3fcc","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-08-10 17:28:57.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"aeff5f828a2b369a58fade99f1c804b66b41cc75","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-10 17:50:41.000000000","message":"Patch Set 23: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/23/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e8c19fdf514c40978444550701b66cc2/index.html : SUCCESS in 11m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/23/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c239bba3e67e4eed91d32addea655c45/index.html : SUCCESS in 12m 38s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/23/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3257bd4af02e479787f23c91d6d6bb51/index.html : SUCCESS in 12m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/23/check/real-db-upgrade_nova_percona_user_001:th-percona/2ca217e072404a21ba0533407d7166ca/index.html : SUCCESS in 13m 26s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/23/check/real-db-upgrade_nova_mysql_user_002:th-mysql/23fa80d19e944e82964128523825c339/index.html : SUCCESS in 13m 51s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/23/check/real-db-upgrade_nova_percona_user_002:th-percona/a697b454b09341a9a65208f0670fd243/index.html : SUCCESS in 11m 33s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":23},{"id":"606e18c8d19590823bc6493ab7d5c9af5fa71dcd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-10 18:23:06.000000000","message":"Patch Set 23: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/23/7726/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":23},{"id":"bfafa445774bccbb3ccade3e714b83943d950ba7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-08-10 18:34:09.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/187857/23/ : FAILURE in 8m 01s","accounts_in_message":[],"_revision_number":23},{"id":"ab5a66ce40b038cfe02cceb6d236e75d0d2ac9bc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-10 18:40:45.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/23/check/dsvm-tempest-xen/9c9f5cf : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":23},{"id":"d0ef767b77b8ac91a6cead8935a01e75134a029f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-10 18:42:41.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/23/check/gate-nova-pep8/c4657c8/ : SUCCESS in 8m 57s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/23/check/gate-nova-docs/cc599ae//doc/build/html/ : SUCCESS in 13m 43s\n- gate-nova-python27 http://logs.openstack.org/57/187857/23/check/gate-nova-python27/03e3965/ : SUCCESS in 12m 11s\n- gate-nova-python34 http://logs.openstack.org/57/187857/23/check/gate-nova-python34/1930be5/ : SUCCESS in 12m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-full/7ff398e/ : SUCCESS in 39m 11s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-postgres-full/3f955e1/ : SUCCESS in 58m 36s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-neutron-full/d4c63f2/ : SUCCESS in 1h 10m 07s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/23/check/gate-grenade-dsvm/3a332c9/ : SUCCESS in 55m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-large-ops/e9c170d/ : SUCCESS in 35m 04s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-neutron-large-ops/c2b9395/ : SUCCESS in 36m 14s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/23/check/gate-devstack-dsvm-cells/fdfe27f/ : SUCCESS in 28m 55s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/23/check/gate-nova-tox-functional/6718a37/ : SUCCESS in 11m 38s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/23/check/gate-grenade-dsvm-partial-ncpu/0b2513b/ : SUCCESS in 52m 01s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-ironic-pxe_ssh/a95fe7b/ : SUCCESS in 51m 51s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-nova-v21-full/bc35fc2/ : SUCCESS in 53m 23s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-cells/cc47639/ : SUCCESS in 1h 00m 54s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-full-ceph/06c7345/ : SUCCESS in 52m 07s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-multinode-full/23cedd5/ : SUCCESS in 49m 50s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-neutron-multinode-full/baa1401/ : SUCCESS in 1h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f9c69ed569a67d80f1062b79e0fe95a841aa73f7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-10 19:15:26.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/187857/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"b3d43929aee039bbc5aebce5fdb785a309893cd9","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-08-10 20:32:59.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/23/check/dsvm-tempest-cinder/572fc63/ : SUCCESS in 34m 04s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"449cd8ec0fa3eeddda95093965c8721cd13c9dee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-10 22:46:23.000000000","message":"Patch Set 23:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/23/check-ibm-tempest-dsvm-full/1a30ea4 : SUCCESS in 52m 50s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/23/check-ibm-tempest-dsvm-postgres-full/99d517a : SUCCESS in 51m 24s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"4f0c1cf32db72ffc2ac554ce910d9c3b1473d9cf","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-12 13:43:07.000000000","message":"Patch Set 23: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/23/7968/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":23},{"id":"9b93a17e834bb14560a1641966c17a07be94f823","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-08-13 07:42:43.000000000","message":"Patch Set 23:\n\nBuild failed. Logs: http://208.91.1.172/logs/187857/23 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"d6600c857cc4ab0b4a3b5748714a7aa628fc546e","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2015-08-22 11:10:02.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"e9eddb3aeca797f3878c9863b4b21ca4419574c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-22 12:29:57.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/23/check/gate-nova-pep8/447af18/ : SUCCESS in 8m 57s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/23/check/gate-nova-docs/d0ab374//doc/build/html/ : SUCCESS in 8m 47s\n- gate-nova-python27 http://logs.openstack.org/57/187857/23/check/gate-nova-python27/d16499c/ : SUCCESS in 11m 22s\n- gate-nova-python34 http://logs.openstack.org/57/187857/23/check/gate-nova-python34/3d7d737/ : SUCCESS in 13m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-full/686c006/ : SUCCESS in 53m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-postgres-full/1816c7e/ : SUCCESS in 44m 58s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-neutron-full/b0ca2d2/ : SUCCESS in 1h 11m 40s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/23/check/gate-grenade-dsvm/f2933ea/ : SUCCESS in 47m 42s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-large-ops/0bd4350/ : SUCCESS in 25m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-neutron-large-ops/8ac0dea/ : SUCCESS in 33m 16s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/23/check/gate-devstack-dsvm-cells/a4d20f7/ : SUCCESS in 26m 52s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/23/check/gate-nova-tox-functional/70610b6/ : FAILURE in 20m 52s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/23/check/gate-grenade-dsvm-partial-ncpu/23bc5a2/ : SUCCESS in 39m 32s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-ironic-pxe_ssh/334c5a5/ : SUCCESS in 41m 46s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-nova-v21-full/f33d282/ : SUCCESS in 44m 10s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-cells/738ba5d/ : SUCCESS in 47m 30s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-full-ceph/14f6c06/ : SUCCESS in 45m 42s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/23/check/gate-tempest-dsvm-multinode-full/e93d197/ : SUCCESS in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"35c712db072158a0b6762437a23c538fd38d4d37","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-08-24 17:21:44.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"8cf917062db428e1bd07e7bc155300724de54bae","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-24 17:22:52.000000000","message":"Patch Set 24:\n\n* pci-test http://52.27.155.124/187857/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"4ce02c0ea4cb1a5899281e7e027b545bb866d926","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-24 17:41:12.000000000","message":"Patch Set 24: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/24/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a632493d7fbd44dabf59dd376a9d1904/index.html : SUCCESS in 14m 05s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/24/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0f562913f0944105b3da0f0d3217768e/index.html : SUCCESS in 15m 33s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/24/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/433dd4aa49954274bac5d964e3e7866e/index.html : SUCCESS in 14m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/24/check/real-db-upgrade_nova_percona_user_001:th-percona/ae1c00e087864f1b8fddba3ca19b5553/index.html : SUCCESS in 12m 49s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/24/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a86209fba13d4abfabe27b89ad025fc3/index.html : SUCCESS in 14m 28s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/24/check/real-db-upgrade_nova_percona_user_002:th-percona/ff02ca1323c54ec59e33d0dee3dcf93a/index.html : SUCCESS in 13m 56s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":24},{"id":"68a0ffebe2641e948e712ef7eda7b51a5f19db26","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-24 18:41:03.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/24/check/dsvm-tempest-xen/16421ce : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":24},{"id":"45274f96cb515620c1d978f2b471522a974ca614","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:51:46.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/24/check/gate-nova-pep8/b9967a8/ : SUCCESS in 9m 29s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/24/check/gate-nova-docs/ac96445//doc/build/html/ : SUCCESS in 14m 05s\n- gate-nova-python27 http://logs.openstack.org/57/187857/24/check/gate-nova-python27/979a1e7/ : SUCCESS in 14m 10s\n- gate-nova-python34 http://logs.openstack.org/57/187857/24/check/gate-nova-python34/2a6349c/ : SUCCESS in 11m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-full/4ee1e72/ : SUCCESS in 53m 13s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-postgres-full/3e0ca02/ : SUCCESS in 59m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-neutron-full/0e0d0fd/ : SUCCESS in 1h 09m 10s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/24/check/gate-grenade-dsvm/d392f86/ : SUCCESS in 37m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-large-ops/0f589c5/ : SUCCESS in 28m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-neutron-large-ops/cdc3d23/ : SUCCESS in 30m 13s\n- gate-devstack-dsvm-cells http://logs.openstack.org/57/187857/24/check/gate-devstack-dsvm-cells/d48a2de/ : SUCCESS in 20m 11s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/24/check/gate-nova-tox-functional/c08bffa/ : SUCCESS in 12m 02s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/24/check/gate-grenade-dsvm-partial-ncpu/cd84f3a/ : SUCCESS in 48m 32s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-ironic-pxe_ssh/a66d6e6/ : SUCCESS in 33m 57s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-nova-v21-full/475003c/ : SUCCESS in 44m 57s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-cells/c6cb2fe/ : SUCCESS in 37m 49s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-full-ceph/189da89/ : SUCCESS in 44m 05s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/24/check/gate-tempest-dsvm-multinode-full/2f83aab/ : FAILURE in 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"4b62f314fc0dfb1e0a91849ce8bdbe8a00831914","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-08-24 19:25:26.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/24/check/dsvm-tempest-cinder/e4322d0/ : FAILURE in 27m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"651508328fd786b2fa917e30e469f83b0ad4579a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-24 19:25:46.000000000","message":"Patch Set 24:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/24/check-ibm-tempest-dsvm-full/77ee751 : SUCCESS in 55m 19s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/24/check-ibm-tempest-dsvm-postgres-full/4d44308 : SUCCESS in 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"76ee6787e5d0b1514773aa588eec21623e26c4ba","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-25 03:06:27.000000000","message":"Patch Set 24:\n\n* pci-test http://52.27.155.124/187857/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"dbfc6720108cd4259072337ad547274517074ba1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-08-26 14:15:49.000000000","message":"Patch Set 24: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/24 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/24 : SUCCESS","accounts_in_message":[],"_revision_number":24},{"id":"367d97d8206e6c7ffa2df0f80583c550d786ac75","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-27 13:55:51.000000000","message":"Patch Set 24: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/24/9600/index.html : FAILURE in 01 h 05 m 57 s","accounts_in_message":[],"_revision_number":24},{"id":"693fa9f6ce199fbf5e916b7690d30b26df632d2a","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-04 14:56:56.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9e861e04ae3714e477ab42a9b1fdaf44cfb314a9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-04 15:33:30.000000000","message":"Patch Set 25: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/25/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3693afb10ba041ec863e96f274b01aef/index.html : SUCCESS in 11m 40s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/25/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ba7013a586804fb2b13e1929cb943da4/index.html : SUCCESS in 16m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/25/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/59561a6763e54343b2a1e25f81ca20fc/index.html : SUCCESS in 12m 34s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/25/check/real-db-upgrade_nova_percona_user_001:th-percona/087ae4160f4a470fbf81f839b16f4b46/index.html : SUCCESS in 16m 10s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/25/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1c8d08d39529424da6891cfe938e7c07/index.html : SUCCESS in 14m 25s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/25/check/real-db-upgrade_nova_percona_user_002:th-percona/c78886132bf942dca8fc35e13d27ddf2/index.html : SUCCESS in 14m 25s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":25},{"id":"614f552d61279ab9550a891a28b31cf92990dd58","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-04 15:48:04.000000000","message":"Patch Set 25:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/25/check-ibm-tempest-dsvm-full/74173cf : SUCCESS in 49m 56s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/25/check-ibm-tempest-dsvm-postgres-full/393746d : SUCCESS in 47m 08s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"a6fbe73b1c3a42eecb6e63f6b9cd8d65efb05016","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-04 15:57:52.000000000","message":"Patch Set 25:\n\n* pci-test http://52.27.155.124/187857/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"acd45fad087a7632da6ad5279fbd4862d68b179c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-04 16:10:50.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/25/check/dsvm-tempest-xen/fae5237 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":25},{"id":"e5e68c909152e95d239f65dc2a55851bc1e1089c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-04 16:12:47.000000000","message":"Patch Set 25: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/25/10976/index.html : SUCCESS in 01 h 12 m 20 s","accounts_in_message":[],"_revision_number":25},{"id":"bef3de85829c64cb2864f26991acddfa50f8b1f2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-04 16:34:32.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/25/check/gate-nova-pep8/6b3bc10/ : SUCCESS in 8m 45s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/25/check/gate-nova-docs/c9d324b//doc/build/html/ : SUCCESS in 11m 02s\n- gate-nova-python27 http://logs.openstack.org/57/187857/25/check/gate-nova-python27/0e784f7/ : SUCCESS in 14m 42s\n- gate-nova-python34 http://logs.openstack.org/57/187857/25/check/gate-nova-python34/5c5301b/ : SUCCESS in 13m 24s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-full/5828b0d/ : SUCCESS in 43m 35s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-postgres-full/3c00840/ : SUCCESS in 44m 53s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-neutron-full/840b255/ : FAILURE in 59m 03s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/25/check/gate-grenade-dsvm/c41d257/ : SUCCESS in 40m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-large-ops/03b752f/ : SUCCESS in 30m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-neutron-large-ops/dd00dd2/ : SUCCESS in 27m 26s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/25/check/gate-nova-tox-functional/00243a1/ : SUCCESS in 16m 37s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/25/check/gate-grenade-dsvm-partial-ncpu/675ae21/ : SUCCESS in 49m 05s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/69d5d21/ : SUCCESS in 41m 26s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-nova-v20-api/dffd386/ : SUCCESS in 35m 51s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-nova-v20-api-legacy/043ffb4/ : SUCCESS in 30m 32s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-nova-v21-full/19c4851/ : SUCCESS in 45m 36s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-cells/e95d05a/ : FAILURE in 40m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-full-ceph/4dfab3e/ : SUCCESS in 42m 37s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/25/check/gate-tempest-dsvm-multinode-full/527c912/ : SUCCESS in 53m 48s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"4ab89684fba30292673a35c86beb91fbf8c18608","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-04 18:13:20.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/187857/25/ : FAILURE in 1h 59m 03s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"6b663d8a35fb6ed548836a37d3f31ad12826366d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-05 14:38:20.000000000","message":"Patch Set 25: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/25 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/25 : SUCCESS","accounts_in_message":[],"_revision_number":25},{"id":"3ab6b080286b9ce98b4d5e30f28813a2ac76053a","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-08 23:09:11.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"4ad01b3e413026f522ba1462160e75941064f4f7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-08 23:30:29.000000000","message":"Patch Set 26:\n\n* pci-test http://52.27.155.124/187857/26 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"3483a535ed4ea9a9c7ad5b24e4527d4a91c18670","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-08 23:33:52.000000000","message":"Patch Set 26: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/26/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7c9509fd4a99405aa4e66ec7ee172c84/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/26/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e0e6721173f14c759a7878d57aff6516/index.html : SUCCESS in 16m 28s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/26/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2a03223cdda248129527270c6ad84349/index.html : SUCCESS in 12m 35s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/26/check/real-db-upgrade_nova_percona_user_001:th-percona/072b3e7212d94fa4ae312849fd720364/index.html : SUCCESS in 15m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/26/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3ee5faf2cb3743d698ba0079be178ad1/index.html : SUCCESS in 13m 52s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/26/check/real-db-upgrade_nova_percona_user_002:th-percona/65aeda5b4c2f45a789f58e87c4afe0b7/index.html : SUCCESS in 13m 01s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":26},{"id":"de3c3fd14504818c299add07d9dd544efa672567","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-08 23:44:18.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/187857/26/ : FAILURE in 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"552ba76d26cbecee63be04e360283159abbbe1fe","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-09 00:12:27.000000000","message":"Patch Set 26: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/26/11277/index.html : SUCCESS in 01 h 00 m 48 s","accounts_in_message":[],"_revision_number":26},{"id":"15a1d43c4ae91d5e7e9b93d5e30c57b6d4d899f1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-09 00:20:04.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/26/check/dsvm-tempest-xen/377dee6 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":26},{"id":"6acbc6483171a0d621b408b48981d89878cbe165","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-09 00:21:15.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/26/check/gate-nova-pep8/28f869d/ : SUCCESS in 6m 55s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/26/check/gate-nova-docs/e22d372//doc/build/html/ : SUCCESS in 11m 16s\n- gate-nova-python27 http://logs.openstack.org/57/187857/26/check/gate-nova-python27/0f0b5ce/ : SUCCESS in 16m 21s\n- gate-nova-python34 http://logs.openstack.org/57/187857/26/check/gate-nova-python34/026a7db/ : SUCCESS in 12m 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-full/c7ebb1d/ : SUCCESS in 56m 32s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-postgres-full/7d9478d/ : SUCCESS in 52m 11s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-neutron-full/1a4f492/ : SUCCESS in 1h 10m 36s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/26/check/gate-grenade-dsvm/87687e6/ : SUCCESS in 46m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-large-ops/0f234f4/ : SUCCESS in 35m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-neutron-large-ops/30abc5e/ : SUCCESS in 34m 20s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/26/check/gate-nova-tox-functional/39184b6/ : SUCCESS in 16m 06s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/26/check/gate-grenade-dsvm-partial-ncpu/2020565/ : SUCCESS in 46m 18s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/cfd57de/ : SUCCESS in 46m 47s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-nova-v20-api/a10b78a/ : SUCCESS in 45m 48s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-nova-v20-api-legacy/0d70e52/ : SUCCESS in 38m 38s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-nova-v21-full/d1e3720/ : SUCCESS in 53m 57s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-cells/5c38478/ : SUCCESS in 50m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-full-ceph/acadd27/ : SUCCESS in 59m 59s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/26/check/gate-tempest-dsvm-multinode-full/2cf50a6/ : SUCCESS in 55m 19s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"266b763859b9f23ca2ee9afab31bf1902cb06566","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-09 00:57:22.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/57/187857/26/check/check-dsvm-vz7-tempest-exe-minimal/bb0c8df/ : FAILURE in 35m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"6684cc1da2369d2757c9e27650e227c5a80efaea","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-09 01:04:37.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/26/check/dsvm-tempest-cinder/877f002/ : SUCCESS in 31m 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"8bb6763a93cf6c900b2df3ad202d309398e20b15","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-09 04:11:57.000000000","message":"Patch Set 26:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/26/check-ibm-tempest-dsvm-full/3d357b2 : SUCCESS in 1h 00m 46s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/26/check-ibm-tempest-dsvm-postgres-full/70dd33d : ABORTED in 2h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"4362224404cb155f97a70041d11b1f3c5a7849a9","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-09-09 06:15:16.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"cd2d05fe0a3eaaa1b48f4bd6b6c7bf44477b4b93","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-09 16:37:13.000000000","message":"Patch Set 26: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"32a3c7dc7508554882d2f2fb2fe050e7e87d3e26","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-09 17:55:32.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"3d0da02c74ed3c9fd0238e0eed9f0d14991c1af1","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-09 17:56:29.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"a3ab5863bd220560f4f9bb251a81b697a729841d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-09 17:57:45.000000000","message":"Patch Set 27:\n\n* pci-test http://52.27.155.124/187857/27 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"7ee9f0ecfef3929c0acf5183f47698f36a290d44","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-09 18:15:02.000000000","message":"Patch Set 28:\n\n* pci-test http://52.27.155.124/187857/28 : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"b44aba980e32513b5b55684326bb6a18677e0061","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-09 18:17:57.000000000","message":"Patch Set 28: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/51fbe83afb0047c983d3f3fbb4d7311e/index.html : SUCCESS in 12m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_mysql_user_001:th-mysql/683b451adc8b46b0ab5aa1eb6c7c00f1/index.html : SUCCESS in 15m 48s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/217237244dab4a7dba7afd4e5179dffb/index.html : SUCCESS in 12m 26s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_percona_user_001:th-percona/47fe4821a984441486cf20ca6de856fa/index.html : SUCCESS in 14m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_mysql_user_002:th-mysql/22df9fbfd00648f9b0b06829a42e072b/index.html : SUCCESS in 14m 15s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_percona_user_002:th-percona/dc0836ad2581424687fb533dea5b4853/index.html : SUCCESS in 14m 36s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":28},{"id":"505a188e0750a830b6e50dec426ecbf047d8809d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-09 19:07:59.000000000","message":"Patch Set 28: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/28/11373/index.html : SUCCESS in 01 h 05 m 55 s","accounts_in_message":[],"_revision_number":28},{"id":"f703f03b3cec78da30cac09b3c437cd9e8b7f9c6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-09 19:10:22.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/28/check/gate-nova-pep8/68506c7/ : SUCCESS in 9m 18s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/28/check/gate-nova-docs/8f20882//doc/build/html/ : SUCCESS in 40m 15s\n- gate-nova-python27 http://logs.openstack.org/57/187857/28/check/gate-nova-python27/7095ab5/ : SUCCESS in 15m 35s\n- gate-nova-python34 http://logs.openstack.org/57/187857/28/check/gate-nova-python34/e5e3bff/ : SUCCESS in 13m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-full/a4bd998/ : SUCCESS in 56m 39s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-postgres-full/9d2373a/ : SUCCESS in 1h 12m 06s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-neutron-full/534eae6/ : SUCCESS in 1h 10m 19s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/28/check/gate-grenade-dsvm/526b4fd/ : FAILURE in 32m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-large-ops/ae1e3ac/ : SUCCESS in 34m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-neutron-large-ops/4faf08c/ : SUCCESS in 31m 11s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/28/check/gate-nova-tox-functional/205fe3a/ : SUCCESS in 11m 35s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/28/check/gate-grenade-dsvm-partial-ncpu/c1cfeb3/ : SUCCESS in 51m 24s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/623abde/ : SUCCESS in 40m 18s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v20-api/0e37e0e/ : SUCCESS in 35m 22s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v20-api-legacy/b59754b/ : SUCCESS in 35m 46s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v21-full/368c23b/ : SUCCESS in 40m 49s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-cells/67dda05/ : SUCCESS in 48m 09s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-full-ceph/93b951c/ : SUCCESS in 51m 01s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-multinode-full/e7cacce/ : SUCCESS in 52m 12s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"7b42c41689ddeb02b7709ca1a3ad4b5eb84ca543","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-09 20:09:49.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/187857/28/ : FAILURE in 2h 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"23e82ce26d8cd98cd4ae9c7e172e33ad970dae22","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-09 20:29:12.000000000","message":"Patch Set 28:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/28/check-ibm-tempest-dsvm-full/be2dbc8 : SUCCESS in 1h 22m 00s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/28/check-ibm-tempest-dsvm-postgres-full/2ce7c18 : SUCCESS in 54m 06s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"8e27226391a576aef02894056a9c195a0a2dda45","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-09 20:54:08.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/28/check/dsvm-tempest-xen/421de1b : FAILURE in 2h 55m 32s","accounts_in_message":[],"_revision_number":28},{"id":"86f4a46aff65ee3633fb56cfa0f2ddcd5009c6e6","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-09 22:09:11.000000000","message":"Patch Set 28:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":28},{"id":"cd7ab941e9e89e12b8b38ecce71346bcf25e9c0b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-09 22:26:30.000000000","message":"Patch Set 28:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/be34e317e68746d9b0a8f5dbc69b607e/index.html : SUCCESS in 11m 55s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5461b032aa5b4b7488ca077481403037/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a61efa6451414e11b2c1be7dbdcbc23a/index.html : SUCCESS in 12m 31s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_percona_user_001:th-percona/cbcc3925e42146febb2885222ac07e65/index.html : SUCCESS in 12m 52s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_mysql_user_002:th-mysql/48acd0edf53344fba1012dc7f424c67d/index.html : SUCCESS in 12m 09s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/28/check/real-db-upgrade_nova_percona_user_002:th-percona/2272645ff3f54115a1d0761e231de525/index.html : SUCCESS in 14m 17s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":28},{"id":"fd95ffa4db704985a34cf0339a78ee28199936bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-09 23:22:54.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/28/check/gate-nova-pep8/d800287/ : SUCCESS in 6m 36s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/28/check/gate-nova-docs/460ac57//doc/build/html/ : SUCCESS in 9m 17s\n- gate-nova-python27 http://logs.openstack.org/57/187857/28/check/gate-nova-python27/abd4d40/ : SUCCESS in 13m 56s\n- gate-nova-python34 http://logs.openstack.org/57/187857/28/check/gate-nova-python34/83a5a89/ : SUCCESS in 12m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-full/79f921e/ : SUCCESS in 44m 39s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-postgres-full/b65b8dc/ : SUCCESS in 38m 47s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-neutron-full/ded4212/ : SUCCESS in 1h 12m 39s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/28/check/gate-grenade-dsvm/fad3a63/ : FAILURE in 35m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-large-ops/0857fdc/ : SUCCESS in 34m 40s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-neutron-large-ops/e348d03/ : SUCCESS in 26m 55s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/28/check/gate-nova-tox-functional/6dda11c/ : SUCCESS in 19m 28s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/28/check/gate-grenade-dsvm-partial-ncpu/ce17f6e/ : SUCCESS in 44m 54s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0f54c08/ : SUCCESS in 35m 55s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v20-api/52dd4f1/ : SUCCESS in 39m 58s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v20-api-legacy/5b59e90/ : SUCCESS in 35m 07s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-cells/c9f56df/ : SUCCESS in 51m 58s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-full-ceph/6a89831/ : SUCCESS in 52m 37s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-multinode-full/72c741c/ : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"4d40394725ae6f101233e0c6e6ee7342f0b486d6","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-09-09 23:23:16.000000000","message":"Patch Set 28:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm: https://bugs.launchpad.net/bugs/1446583\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":28},{"id":"07b3b893a9d4d19ecdb6ce2041ad39622c06c065","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-10 00:50:02.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/187857/28/ : FAILURE in 2h 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"a4269602622794bdc452bde61bbd135dd849039f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-10 01:05:46.000000000","message":"Patch Set 28:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/187857/28/check/dsvm-tempest-xen/619a806 : FAILURE in 2h 55m 46s","accounts_in_message":[],"_revision_number":28},{"id":"16da857ffc70ee1dc73ce11637025edd5ce19a0a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-10 14:53:07.000000000","message":"Patch Set 28: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/28 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/28 : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"3e1156ec92cf3c1b481aeb9ab8b78891a60f9205","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-10 21:29:11.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"9395d45d35f95a5830409a84d072892e8fb5ed34","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-10 23:04:50.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/28/check/gate-nova-pep8/a8b6378/ : SUCCESS in 8m 23s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/28/check/gate-nova-docs/535aa9b//doc/build/html/ : SUCCESS in 15m 50s\n- gate-nova-python27 http://logs.openstack.org/57/187857/28/check/gate-nova-python27/5929f6e/ : SUCCESS in 15m 05s\n- gate-nova-python34 http://logs.openstack.org/57/187857/28/check/gate-nova-python34/b8b88be/ : SUCCESS in 10m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-full/9172f21/ : SUCCESS in 42m 39s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-postgres-full/df950c8/ : SUCCESS in 42m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-neutron-full/4f9aa6d/ : SUCCESS in 59m 44s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/28/check/gate-grenade-dsvm/fafb49c/ : SUCCESS in 38m 55s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-large-ops/aefe923/ : SUCCESS in 27m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-neutron-large-ops/2904782/ : SUCCESS in 32m 43s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/28/check/gate-nova-tox-functional/1fed91e/ : SUCCESS in 15m 35s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/28/check/gate-grenade-dsvm-partial-ncpu/03edad7/ : SUCCESS in 45m 11s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8899d60/ : SUCCESS in 42m 36s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v20-api/5f8b4a9/ : SUCCESS in 40m 52s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-nova-v20-api-legacy/b7d2cdc/ : SUCCESS in 38m 58s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-cells/09058cf/ : SUCCESS in 52m 18s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-full-ceph/dc38839/ : SUCCESS in 52m 45s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/28/check/gate-tempest-dsvm-multinode-full/b00a6a4/ : SUCCESS in 1h 03m 02s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"4de2f82780bca79781935220c7936c040d73c4bb","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-11 00:02:27.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/57/187857/28/check/check-dsvm-vz7-tempest-exe-minimal/0951828/ : FAILURE in 41m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":28},{"id":"205dd6fc2953b63b6dba847bb916ab5492dee566","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-11 01:33:24.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/28/check/dsvm-tempest-cinder/bb8f397/ : SUCCESS in 54m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":28},{"id":"f4e21c13c481454aaa06b0656dec386bdda163f4","author":{"_account_id":1107,"name":"Josh Durgin","email":"jdurgin@redhat.com","username":"jdurgin"},"date":"2015-09-14 19:24:38.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"154649801db464f5a5291253cba28afcf583e670","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-15 01:05:45.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"00cde5bc0ea86db6f7b92d236a0749fa4b4a6ad8","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-15 01:38:51.000000000","message":"Patch Set 29: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/29/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/23f4e4aa0e364701b8ace5304a627abc/index.html : SUCCESS in 12m 57s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/29/check/real-db-upgrade_nova_mysql_user_001:th-mysql/78fe650ef3484b97ab291626c14ee73a/index.html : SUCCESS in 15m 40s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/29/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f98431cea3134726a19ab01998c900be/index.html : SUCCESS in 12m 34s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/29/check/real-db-upgrade_nova_percona_user_001:th-percona/881e817af67a478d93e2049055346c26/index.html : SUCCESS in 13m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/57/187857/29/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1e49fcbc9e72484bab31478ed7be6eda/index.html : SUCCESS in 14m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/57/187857/29/check/real-db-upgrade_nova_percona_user_002:th-percona/09f81ba9482d454583c032e7645cd68c/index.html : SUCCESS in 14m 41s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":29},{"id":"4bc05a563c6e5bf25b00ae10e162bab158472558","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-15 02:00:23.000000000","message":"Patch Set 29:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/29/check-ibm-tempest-dsvm-full/785b54b : SUCCESS in 53m 58s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/187857/29/check-ibm-tempest-dsvm-postgres-full/0384eb7 : SUCCESS in 50m 56s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"2a71a0712093b51db128abecb2d43a84a3b5aafc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-15 02:18:52.000000000","message":"Patch Set 29: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/187857/29/11698/index.html : SUCCESS in 01 h 04 m 40 s","accounts_in_message":[],"_revision_number":29},{"id":"07b744e1f70d6959a8e77a1d732aad8357213552","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-15 02:27:50.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/57/187857/29/check/gate-nova-pep8/47a3913/ : SUCCESS in 6m 54s\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/29/check/gate-nova-docs/301561d//doc/build/html/ : SUCCESS in 9m 36s\n- gate-nova-python27 http://logs.openstack.org/57/187857/29/check/gate-nova-python27/f60756b/ : SUCCESS in 11m 22s\n- gate-nova-python34 http://logs.openstack.org/57/187857/29/check/gate-nova-python34/c96dab7/ : SUCCESS in 14m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-full/6fa6cd8/ : SUCCESS in 1h 02m 44s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-postgres-full/7c41541/ : SUCCESS in 51m 08s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-neutron-full/2285821/ : SUCCESS in 1h 20m 52s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/29/check/gate-grenade-dsvm/dc42cf7/ : SUCCESS in 45m 24s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-large-ops/9f79a7f/ : SUCCESS in 37m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-neutron-large-ops/f38080b/ : SUCCESS in 34m 37s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/29/check/gate-nova-tox-functional/b007eef/ : SUCCESS in 12m 48s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/29/check/gate-grenade-dsvm-partial-ncpu/8298fe8/ : SUCCESS in 47m 22s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/fe401ef/ : SUCCESS in 49m 38s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-nova-v20-api/7ed2269/ : SUCCESS in 29m 32s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-nova-v20-api-legacy/ceb1a9d/ : SUCCESS in 32m 16s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-cells/991e159/ : SUCCESS in 42m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-full-ceph/49ed400/ : SUCCESS in 44m 14s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/57/187857/29/check/gate-tempest-dsvm-multinode-full/c814f86/ : SUCCESS in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"a29466ce506cb61468cc4f74af92e8d70e1cf989","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-15 02:31:14.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/187857/29/Nova-ML2-Sriov : FAILURE in 16m 02s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":29},{"id":"cec8639acb2099e1ada3fc1a404af3a2396e14b1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-15 03:05:52.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/57/187857/29/check/check-dsvm-vz7-tempest-exe-minimal/7d00cf1/ : FAILURE in 35m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":29},{"id":"29b1796873e46b8c1b108193b5b136b49f63a479","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-15 03:16:45.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/57/187857/29/check/dsvm-tempest-cinder/86e56a2/ : FAILURE in 42m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":29},{"id":"77a6bb83e1c4ca20ab73c783c0f8ee8ca70d684f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-15 04:06:36.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/187857/29/ : FAILURE in 3h 00m 01s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"a991239401160740b5e1150cf25ea328a6998440","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-15 04:36:33.000000000","message":"Patch Set 29: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/187857/29 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/187857/29 : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"38d87560554fcecef895e44a7ae5d2042ec9bfd4","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2015-09-23 01:13:33.000000000","message":"Patch Set 29: Code-Review+2\n\nLooks reasonable to me.","accounts_in_message":[],"_revision_number":29},{"id":"00bea350888da87142ca4eff630dab139018810f","author":{"_account_id":17106,"name":"IanSun","email":"sun.jun@99cloud.net","username":"SunJun"},"date":"2015-09-24 09:23:47.000000000","message":"Patch Set 29: Code-Review+1\n\ngood work","accounts_in_message":[],"_revision_number":29},{"id":"88c18f10c70dc3b806ac7273938362f1ea7593b2","author":{"_account_id":7135,"name":"Liang Chen","email":"liangchen.linux@gmail.com","username":"Liang"},"date":"2015-09-25 06:45:52.000000000","message":"Patch Set 29: Code-Review+1","accounts_in_message":[],"_revision_number":29},{"id":"f743b0f2f47eccba8d6e003923baad89e37909cc","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2015-09-25 13:40:24.000000000","message":"Patch Set 29: Code-Review+1\n\nSeems fair","accounts_in_message":[],"_revision_number":29},{"id":"5dff0d0c238ec9f52ad37fff32c1054885c1d755","author":{"_account_id":642,"name":"Matt Dietz","email":"matthew.dietz@gmail.com","username":"cerberus"},"date":"2015-09-28 14:52:07.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":29},{"id":"a6235037b49d9c120980d7e7157469d06628118f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-28 14:52:15.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":29},{"id":"e33299a279120d74f93bf9755dbe1b8decebed66","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-09-28 15:41:13.000000000","message":"Patch Set 29: Cherry Picked\n\nThis patchset was cherry picked to branch stable/liberty as commit 1a22f272e00820548df219aad3683311095775c9","accounts_in_message":[],"_revision_number":29},{"id":"d49db9146893e7bb50dd1811a91679dfa7fbc4d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-28 17:49:10.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/57/187857/29/gate/gate-nova-docs/011aa5a//doc/build/html/ : SUCCESS in 13m 56s\n- gate-nova-pep8 http://logs.openstack.org/57/187857/29/gate/gate-nova-pep8/ed6720b/ : SUCCESS in 8m 59s\n- gate-nova-python27 http://logs.openstack.org/57/187857/29/gate/gate-nova-python27/7c8a126/ : SUCCESS in 15m 36s\n- gate-nova-python34 http://logs.openstack.org/57/187857/29/gate/gate-nova-python34/2056441/ : SUCCESS in 16m 12s\n- gate-tempest-dsvm-full http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-full/25ad1f1/ : SUCCESS in 1h 03m 32s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-postgres-full/6583b69/ : SUCCESS in 1h 00m 56s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-neutron-full/c0800fa/ : SUCCESS in 1h 18m 49s\n- gate-grenade-dsvm http://logs.openstack.org/57/187857/29/gate/gate-grenade-dsvm/ce60522/ : SUCCESS in 57m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-large-ops/ee52b96/ : SUCCESS in 53m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-neutron-large-ops/ca4f2c8/ : SUCCESS in 42m 04s\n- gate-nova-tox-functional http://logs.openstack.org/57/187857/29/gate/gate-nova-tox-functional/dafaea2/ : SUCCESS in 18m 00s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/57/187857/29/gate/gate-grenade-dsvm-partial-ncpu/4efb5fa/ : SUCCESS in 59m 04s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-full-ceph/0f0c287/ : SUCCESS in 1h 04m 01s\n- gate-tempest-dsvm-cells http://logs.openstack.org/57/187857/29/gate/gate-tempest-dsvm-cells/c17df88/ : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":29},{"id":"b0e6a9d1d523e6d25ba72743c4b582ce3b39f6b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-28 17:49:14.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":29},{"id":"47d73e0a16d9365484c00a2aff900ccdb0c81f8d","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2016-08-16 14:21:59.000000000","message":"Patch Set 29:\n\nJens, since this patch is also linked to bug 1608934 could add:\n\nCloses-Bug: 1608934","accounts_in_message":[],"_revision_number":29},{"id":"572eeab384d878ac55e38be6cc9d278c73fea360","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2016-08-16 14:22:28.000000000","message":"Patch Set 29:\n\nOops pasted comment on wrong patch sorry","accounts_in_message":[],"_revision_number":29},{"id":"4c950713a322f644fa21d40f74169bd1cf2585a0","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-08-18 17:48:29.000000000","message":"Patch Set 29: Reverted\n\nThis patchset was reverted in change: I3c71f32d1622d23448750375d9e412af60cbc616","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"c56bbc61f44529042ae9126a9a0b62affe0276d0","revisions":{"6e1908962d1f90246bc74b6bdc1f919254b30f53":{"kind":"REWORK","_number":1,"created":"2015-06-03 04:08:55.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/1"}}},"commit":{"parents":[{"commit":"5fd4cb7f5dfd38ab55ddb9a06ff00d6db5cbd208","subject":"Merge \"Make nova-manage handle completely missing flavor information\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5fd4cb7f5dfd38ab55ddb9a06ff00d6db5cbd208"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 23:03:03.000000000","tz":-420},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e1908962d1f90246bc74b6bdc1f919254b30f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e1908962d1f90246bc74b6bdc1f919254b30f53"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5fd4cb7f5dfd38ab55ddb9a06ff00d6db5cbd208","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a01e3e4cb9a450b0b689d5a5824f3e56c998bcfc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2015-06-03 04:09:53.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/2"}}},"commit":{"parents":[{"commit":"b1bc4362041af37cb96567da341aadf0f6846dd1","subject":"Merge \"VMware: use get_datastore_by_ref from oslo.vmware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1bc4362041af37cb96567da341aadf0f6846dd1"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-03 04:09:52.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a01e3e4cb9a450b0b689d5a5824f3e56c998bcfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a01e3e4cb9a450b0b689d5a5824f3e56c998bcfc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1bc4362041af37cb96567da341aadf0f6846dd1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6ae6e8e3d74d377dcf1a0b58d5cb91c7b8e31a53":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2015-06-03 18:53:04.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/3"}}},"commit":{"parents":[{"commit":"70c6b57751cf67f5a2be80d7b2d2123c826a8a68","subject":"Merge \"Make InstanceExternalEvent use an Enum for status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70c6b57751cf67f5a2be80d7b2d2123c826a8a68"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-03 18:19:53.000000000","tz":-420},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ae6e8e3d74d377dcf1a0b58d5cb91c7b8e31a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ae6e8e3d74d377dcf1a0b58d5cb91c7b8e31a53"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"70c6b57751cf67f5a2be80d7b2d2123c826a8a68","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b554b95353dde53959f87c0f8a152241d1220f97":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2015-06-08 22:44:01.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/4"}}},"commit":{"parents":[{"commit":"770a897179f09546b32763697319d49be1697be1","subject":"Merge \"Begin the transition to an explicit object registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/770a897179f09546b32763697319d49be1697be1"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-08 22:44:00.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b554b95353dde53959f87c0f8a152241d1220f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b554b95353dde53959f87c0f8a152241d1220f97"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"770a897179f09546b32763697319d49be1697be1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b2473bfd927fb6e5c1dddcd94c9e20f0ab934f6d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2015-06-16 17:10:45.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/5"}}},"commit":{"parents":[{"commit":"605ef40d6caa6c9c48e9630ddbe8084491b72d63","subject":"Merge \"Add Host Mapping table to API Database\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/605ef40d6caa6c9c48e9630ddbe8084491b72d63"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-16 17:10:44.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2473bfd927fb6e5c1dddcd94c9e20f0ab934f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2473bfd927fb6e5c1dddcd94c9e20f0ab934f6d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"605ef40d6caa6c9c48e9630ddbe8084491b72d63","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bf80e215f1621f3c17f9d95bdc97c0a81a0aa847":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2015-06-17 17:51:31.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/6"}}},"commit":{"parents":[{"commit":"a4b4a79549a9f9e2f0a2b575dd8d3e2d74623735","subject":"Merge \"Add a finish log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4b4a79549a9f9e2f0a2b575dd8d3e2d74623735"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-17 17:51:29.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf80e215f1621f3c17f9d95bdc97c0a81a0aa847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf80e215f1621f3c17f9d95bdc97c0a81a0aa847"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a4b4a79549a9f9e2f0a2b575dd8d3e2d74623735","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"95ed2285ab3611795d0a39c5ad3e68eaf75c10c6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2015-06-18 17:37:06.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/7"}}},"commit":{"parents":[{"commit":"875c0a84b798dac558f21c2165d63488347c3054","subject":"Merge \"Fix wrong check when use image in local\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/875c0a84b798dac558f21c2165d63488347c3054"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-18 17:37:05.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95ed2285ab3611795d0a39c5ad3e68eaf75c10c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95ed2285ab3611795d0a39c5ad3e68eaf75c10c6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"875c0a84b798dac558f21c2165d63488347c3054","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2c9d22d4ad738eef642fdbd6c7cd7012d65d9f91":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2015-06-19 17:42:08.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/8"}}},"commit":{"parents":[{"commit":"f729a1e815f9b7fcfc91fb2b3c2cabbe3f4af7d5","subject":"Merge \"The devref for Nova stable API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f729a1e815f9b7fcfc91fb2b3c2cabbe3f4af7d5"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-19 17:42:06.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c9d22d4ad738eef642fdbd6c7cd7012d65d9f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c9d22d4ad738eef642fdbd6c7cd7012d65d9f91"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f729a1e815f9b7fcfc91fb2b3c2cabbe3f4af7d5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"44ac868a19a55df50267a9fb5d5c5dbcbc2822ce":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2015-06-22 16:51:31.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/9"}}},"commit":{"parents":[{"commit":"cdd56e16fdae4d33cf282d7b36835d59869f55ae","subject":"Merge \"xenapi: remove bittorrent entry point lookup code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdd56e16fdae4d33cf282d7b36835d59869f55ae"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-22 16:51:29.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44ac868a19a55df50267a9fb5d5c5dbcbc2822ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44ac868a19a55df50267a9fb5d5c5dbcbc2822ce"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cdd56e16fdae4d33cf282d7b36835d59869f55ae","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"60ed9a12a4894f17e72822e5c5390a393c2b59f3":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2015-07-08 21:08:31.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/10"}}},"commit":{"parents":[{"commit":"47b233f988bddd1272e4a1bc02be97f15701a8ca","subject":"Merge \"devref: virtual machine states and transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/47b233f988bddd1272e4a1bc02be97f15701a8ca"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-08 21:08:30.000000000","tz":0},"subject":"imagebackend: Split out resize_image logic from create_image","message":"imagebackend: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60ed9a12a4894f17e72822e5c5390a393c2b59f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60ed9a12a4894f17e72822e5c5390a393c2b59f3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"47b233f988bddd1272e4a1bc02be97f15701a8ca","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1bd6ba2877c5507694f8f72e65da224a33fda349":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2015-07-09 18:49:39.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/11"}}},"commit":{"parents":[{"commit":"47b233f988bddd1272e4a1bc02be97f15701a8ca","subject":"Merge \"devref: virtual machine states and transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/47b233f988bddd1272e4a1bc02be97f15701a8ca"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-09 18:49:39.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1bd6ba2877c5507694f8f72e65da224a33fda349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1bd6ba2877c5507694f8f72e65da224a33fda349"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"47b233f988bddd1272e4a1bc02be97f15701a8ca","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ce5d9659a0a1645c6e7a9b1e6710273ba11c591f":{"kind":"REWORK","_number":12,"created":"2015-07-09 19:26:04.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/12"}}},"commit":{"parents":[{"commit":"f93ce86d4ce5c68e3596a7981992e2851549f470","subject":"Merge \"Cells: cell scheduler anti-affinity filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f93ce86d4ce5c68e3596a7981992e2851549f470"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-09 19:25:53.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce5d9659a0a1645c6e7a9b1e6710273ba11c591f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce5d9659a0a1645c6e7a9b1e6710273ba11c591f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f93ce86d4ce5c68e3596a7981992e2851549f470","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"beab13aed72c7e0e3596cce057eb9b030fd8c3fd":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2015-07-21 14:47:12.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/13"}}},"commit":{"parents":[{"commit":"4454cf1d268b12d3c3552b374ca46b0c6ef47bbc","subject":"Do not import using oslo namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4454cf1d268b12d3c3552b374ca46b0c6ef47bbc"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-21 14:47:11.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/beab13aed72c7e0e3596cce057eb9b030fd8c3fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/beab13aed72c7e0e3596cce057eb9b030fd8c3fd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4454cf1d268b12d3c3552b374ca46b0c6ef47bbc","is_merged_in_target_branch":true,"change_id":"I9090aa92947fc11c10f46404fe0c9bf0b9ecc2df","change_number":203786,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bf3e7b80287d75f122d3049288e199a5420161f4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2015-07-22 03:25:14.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/14"}}},"commit":{"parents":[{"commit":"00f85052b53429d5ce6080fc4214a97584374a57","subject":"Merge \"Fix libguestfs failure in test_can_resize_need_fs_type_specified\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00f85052b53429d5ce6080fc4214a97584374a57"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-22 03:25:13.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf3e7b80287d75f122d3049288e199a5420161f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf3e7b80287d75f122d3049288e199a5420161f4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"00f85052b53429d5ce6080fc4214a97584374a57","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2d679bac42b8cd856537821a7895aabf88560698":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2015-07-22 13:04:11.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/15"}}},"commit":{"parents":[{"commit":"6a24bbeecd8a6d6d3135a10f4917b071896d14ee","subject":"Merge \"Switch to oslo.service library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a24bbeecd8a6d6d3135a10f4917b071896d14ee"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-22 13:04:10.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d679bac42b8cd856537821a7895aabf88560698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d679bac42b8cd856537821a7895aabf88560698"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6a24bbeecd8a6d6d3135a10f4917b071896d14ee","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1c660aba586bfcc8a592ec86825b91d2d8383e0f":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2015-07-22 18:53:26.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/16"}}},"commit":{"parents":[{"commit":"0adaed21c856d39a2e1bd4b0bb361f3d3da221ac","subject":"Merge \"Remove db layer hard-code permission checks for archive_deleted_rows*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0adaed21c856d39a2e1bd4b0bb361f3d3da221ac"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-22 18:53:25.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c660aba586bfcc8a592ec86825b91d2d8383e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c660aba586bfcc8a592ec86825b91d2d8383e0f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0adaed21c856d39a2e1bd4b0bb361f3d3da221ac","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d99a97b10620ae7a72e4d2ef97f0edab0cbc823c":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2015-07-23 14:17:33.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/17"}}},"commit":{"parents":[{"commit":"9c5f39d45fefbb16cc1c42b181a9dc24240fccaa","subject":"Merge \"Revert \"Add config drive support for Virtuozzo containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9c5f39d45fefbb16cc1c42b181a9dc24240fccaa"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-23 14:17:32.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d99a97b10620ae7a72e4d2ef97f0edab0cbc823c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d99a97b10620ae7a72e4d2ef97f0edab0cbc823c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9c5f39d45fefbb16cc1c42b181a9dc24240fccaa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1e2cc8ed97ffd385e8f0c18ef1a1a28c64d238be":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2015-07-24 02:09:24.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/18"}}},"commit":{"parents":[{"commit":"f5abf217efa71947c19132fc15f2f57937e0e655","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5abf217efa71947c19132fc15f2f57937e0e655"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-24 02:09:23.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e2cc8ed97ffd385e8f0c18ef1a1a28c64d238be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e2cc8ed97ffd385e8f0c18ef1a1a28c64d238be"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5abf217efa71947c19132fc15f2f57937e0e655","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e561182e1985eb753490057720258d68602d81da":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2015-07-24 16:50:06.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/19"}}},"commit":{"parents":[{"commit":"0d696adcea400da3435382a49a632f23283242f3","subject":"Merge \"libvirt: make unit tests concise by setup guest object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d696adcea400da3435382a49a632f23283242f3"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-24 16:50:05.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e561182e1985eb753490057720258d68602d81da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e561182e1985eb753490057720258d68602d81da"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0d696adcea400da3435382a49a632f23283242f3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"afbb8370a54967afaaf481510a5ce503d449d802":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2015-07-27 01:28:03.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/20"}}},"commit":{"parents":[{"commit":"183cd88cb2f9781b53f71b6b161df401c286c9ff","subject":"Merge \"remove _rescan_iscsi from disconnect_volume_multipath_iscsi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/183cd88cb2f9781b53f71b6b161df401c286c9ff"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-27 01:28:02.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afbb8370a54967afaaf481510a5ce503d449d802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afbb8370a54967afaaf481510a5ce503d449d802"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"183cd88cb2f9781b53f71b6b161df401c286c9ff","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ce66752561ee0467ee44f21d9fd60adb655b3c10":{"kind":"REWORK","_number":21,"created":"2015-07-28 00:12:21.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/21"}}},"commit":{"parents":[{"commit":"183cd88cb2f9781b53f71b6b161df401c286c9ff","subject":"Merge \"remove _rescan_iscsi from disconnect_volume_multipath_iscsi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/183cd88cb2f9781b53f71b6b161df401c286c9ff"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-28 00:11:34.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce66752561ee0467ee44f21d9fd60adb655b3c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce66752561ee0467ee44f21d9fd60adb655b3c10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"183cd88cb2f9781b53f71b6b161df401c286c9ff","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7b92a2d6356d321088ad1673855eb5d5ace05315":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2015-07-28 16:06:52.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/22"}}},"commit":{"parents":[{"commit":"a43a1a37030736324e4acddeaaf0a143243bb6d3","subject":"Merge \"Switch to using os-brick\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a43a1a37030736324e4acddeaaf0a143243bb6d3"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-07-28 16:06:06.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b92a2d6356d321088ad1673855eb5d5ace05315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b92a2d6356d321088ad1673855eb5d5ace05315"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a43a1a37030736324e4acddeaaf0a143243bb6d3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a3c55a9579ee9603334339d294eaf543da8f8c3d":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2015-08-10 17:28:57.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/23"}}},"commit":{"parents":[{"commit":"e239060e41e208674701c80b15b8e754ce1c98cd","subject":"Merge \"Include DiskFilter in the default list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e239060e41e208674701c80b15b8e754ce1c98cd"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-08-10 17:28:55.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3c55a9579ee9603334339d294eaf543da8f8c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3c55a9579ee9603334339d294eaf543da8f8c3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e239060e41e208674701c80b15b8e754ce1c98cd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9641851b9ea313589084abaee1442e11e1779d3b":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2015-08-24 17:21:44.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/24"}}},"commit":{"parents":[{"commit":"46568ccc097d715c270cfc08b31c48c4670faae8","subject":"Merge \"libvirt: only get bdm in _create_domain_setup_lxc if booted from volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46568ccc097d715c270cfc08b31c48c4670faae8"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-08-24 17:21:42.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9641851b9ea313589084abaee1442e11e1779d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9641851b9ea313589084abaee1442e11e1779d3b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"46568ccc097d715c270cfc08b31c48c4670faae8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"010b2b05927dac89465ef4735961d1f5aafac689":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2015-09-04 14:56:56.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/25"}}},"commit":{"parents":[{"commit":"02174c35523acf8e98eac32ec3f5ae41e1dc5dfc","subject":"Merge \"Add migration policy to upgrades devref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02174c35523acf8e98eac32ec3f5ae41e1dc5dfc"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-09-04 14:55:01.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/010b2b05927dac89465ef4735961d1f5aafac689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/010b2b05927dac89465ef4735961d1f5aafac689"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"02174c35523acf8e98eac32ec3f5ae41e1dc5dfc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"606ef9859897e2e23ae65f7320dd5f240a609c14":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2015-09-08 23:09:11.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/26"}}},"commit":{"parents":[{"commit":"c510c8341733bd11d5e74eb3810b77ffaf1fa9b3","subject":"Merge \"Fix order of arguments in assertEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c510c8341733bd11d5e74eb3810b77ffaf1fa9b3"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-09-08 22:13:10.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/606ef9859897e2e23ae65f7320dd5f240a609c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/606ef9859897e2e23ae65f7320dd5f240a609c14"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c510c8341733bd11d5e74eb3810b77ffaf1fa9b3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0d45a6badea5f599ae7074a43ff985cf771ec29d":{"kind":"REWORK","_number":27,"created":"2015-09-09 17:55:32.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/27"}}},"commit":{"parents":[{"commit":"c510c8341733bd11d5e74eb3810b77ffaf1fa9b3","subject":"Merge \"Fix order of arguments in assertEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c510c8341733bd11d5e74eb3810b77ffaf1fa9b3"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-09-09 17:54:19.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work when appropriate.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d45a6badea5f599ae7074a43ff985cf771ec29d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d45a6badea5f599ae7074a43ff985cf771ec29d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c510c8341733bd11d5e74eb3810b77ffaf1fa9b3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fd72f0b905ffc691749b40b15dd80599fd8a73f7":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2015-09-09 17:56:29.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/28"}}},"commit":{"parents":[{"commit":"915e0d18b9cb9d8ff6d295fad19482d178d7655a","subject":"Merge \"Add os-brick\u0027s scsi_id command to rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/915e0d18b9cb9d8ff6d295fad19482d178d7655a"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-09-09 17:56:21.000000000","tz":0},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work when appropriate.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd72f0b905ffc691749b40b15dd80599fd8a73f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd72f0b905ffc691749b40b15dd80599fd8a73f7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"915e0d18b9cb9d8ff6d295fad19482d178d7655a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c56bbc61f44529042ae9126a9a0b62affe0276d0":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2015-09-15 01:05:45.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/57/187857/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/187857/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/187857/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/187857/29"}}},"commit":{"parents":[{"commit":"411769be8fde6311e1ffac066a47ce65bb4a7e08","subject":"Merge \"Creating instance fail when inject ssh key in cells mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/411769be8fde6311e1ffac066a47ce65bb4a7e08"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-06-02 21:59:42.000000000","tz":-420},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2015-09-15 01:05:28.000000000","tz":-420},"subject":"libvirt: Split out resize_image logic from create_image","message":"libvirt: Split out resize_image logic from create_image\n\nRather than rely on the dubiously-named create_image() method to\nhandle resizing existing images, let the imagebackend caching layer\nfigure out if the resize needs to be done, and call resize_image()\nto do the work when appropriate.\n\nPartial-Bug: 1369465\nChange-Id: I28ba1970317f3d5691d05283796d4345ff38a804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c56bbc61f44529042ae9126a9a0b62affe0276d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c56bbc61f44529042ae9126a9a0b62affe0276d0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"411769be8fde6311e1ffac066a47ce65bb4a7e08","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
