)]}'
{"id":"openstack%2Fnova~193135","triplet_id":"openstack%2Fnova~master~Ib78fcd7c54c966dcbc143165a262e582bbda0901","project":"openstack/nova","branch":"master","topic":"bp/volume-multi-attach","hashtags":[],"change_id":"Ib78fcd7c54c966dcbc143165a262e582bbda0901","subject":"libvirt: Allow multiple volume attachments","status":"ABANDONED","created":"2015-06-18 13:45:12.000000000","updated":"2016-02-02 15:21:37.000000000","total_comment_count":38,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba0e756c930c317ab4188cbf67738c72f1a3ea73","_number":193135,"virtual_id_number":193135,"owner":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"date":"2016-01-21 19:12:24.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-01-27 19:37:03.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-01-21 19:22:55.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},{"value":0,"date":"2016-01-21 19:41:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":1,"date":"2016-01-28 23:18:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"date":"2016-01-23 22:10:39.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2016-01-21 19:27:17.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-01-21 22:58:33.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":1,"date":"2016-01-21 19:29:27.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"date":"2016-01-22 06:49:12.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},{"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-06-18 13:47:56.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2015-07-07 11:43:41.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2015-08-04 12:31:53.000000000","updated_by":{"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"},"reviewer":{"_account_id":8190,"name":"Drew Thorstensen (thorst)","email":"thorst@us.ibm.com","username":"thorst"},"state":"REVIEWER"},{"updated":"2015-12-15 22:46:28.000000000","updated_by":{"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},"reviewer":{"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},"state":"REVIEWER"},{"updated":"2015-12-17 12:25:50.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2015-12-17 20:13:48.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2015-12-23 06:21:57.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-05 18:17:29.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"},{"updated":"2016-01-06 10:58:14.000000000","updated_by":{"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},"reviewer":{"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},"state":"REVIEWER"},{"updated":"2016-01-13 02:05:30.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2016-01-19 01:46:39.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-19 07:41:16.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2016-01-19 09:38:07.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2016-01-20 07:11:29.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"},{"updated":"2016-01-21 19:12:24.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-21 19:22:55.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-21 19:27:17.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-21 19:29:27.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-21 19:41:13.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-21 22:58:33.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-22 06:49:12.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-23 22:10:39.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-27 19:37:03.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-01-28 23:18:52.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c0009000732870c5000d85f0be7eaca0115f94d","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-06-18 13:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ee2026c094104eaaa3b0da3e6ff509327bb8467","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-06-18 13:52:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4f9f7b5bd8dcb0df5b2f8940ea46204dcb6f64b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-06-18 14:55:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/2/Nova-ML2-Sriov : SUCCESS in 44m 43s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"db2a3129b2db99fbfb317850d491c9517844ee95","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-18 15:06:36.000000000","message":"Patch Set 2: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/2/1604/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"d9b70789d1cde214afbab4dd53423773dad2e7fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-18 15:07:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/2/check/gate-nova-pep8/7cab312/ : FAILURE in 8m 33s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/2/check/gate-nova-docs/ae1c7e2//doc/build/html/ : SUCCESS in 9m 26s\n- gate-nova-python27 http://logs.openstack.org/35/193135/2/check/gate-nova-python27/7199353/ : SUCCESS in 13m 57s\n- check-tempest-dsvm-full http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-full/17af4d4/ : FAILURE in 1h 10m 36s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-postgres-full/38e47af/ : FAILURE in 1h 01m 33s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-neutron-full/c59fad6/ : FAILURE in 1h 04m 28s\n- check-grenade-dsvm http://logs.openstack.org/35/193135/2/check/check-grenade-dsvm/de6b132/ : FAILURE in 33m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/2/check/gate-tempest-dsvm-large-ops/9718b88/ : SUCCESS in 22m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/2/check/gate-tempest-dsvm-neutron-large-ops/ebdbc0b/ : SUCCESS in 25m 40s\n- check-devstack-dsvm-cells http://logs.openstack.org/35/193135/2/check/check-devstack-dsvm-cells/daced9d/ : FAILURE in 22m 12s\n- gate-nova-python34 http://logs.openstack.org/35/193135/2/check/gate-nova-python34/015928f/ : SUCCESS in 9m 30s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/2/check/gate-nova-tox-functional/482fd41/ : SUCCESS in 11m 56s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/2/check/check-grenade-dsvm-partial-ncpu/6f2d8b6/ : SUCCESS in 39m 47s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-ironic-pxe_ssh/c5c41b5/ : SUCCESS in 33m 36s\n- check-tempest-dsvm-docker http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-docker/a1aa816/ : SUCCESS in 34m 58s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-nova-v21-full/0ae384c/ : FAILURE in 1h 04m 30s\n- check-tempest-dsvm-cells http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-cells/bfbbbc6/ : FAILURE in 48m 01s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-full-ceph/9698dcc/ : FAILURE in 1h 00m 23s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/2/check/check-tempest-dsvm-multinode-full/da5ddef/ : FAILURE in 1h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a5f77366faa502111d601c2b96bf33ee0072fe9","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-18 15:12:09.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: unrecognized error\n- check-grenade-dsvm: unrecognized error\n- check-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-nova-v21-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- gate-nova-pep8: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"033d42faa85c2e57f58d4054633aa2993cc35b79","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-18 15:45:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/2/check/dsvm-tempest-xen/8c5a5e6 : FAILURE in 1h 43m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f4239d49fa8bff34c877a81db0ab76f236a01f87","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-18 16:19:16.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/2/check-ibm-tempest-dsvm-full/154a28c : SUCCESS in 59m 19s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/2/check-ibm-tempest-dsvm-postgres-full/a237c33 : SUCCESS in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b842c9e216b39de527e1bf3ebda89bc7b76f538","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-18 18:09:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/193135/2/ : FAILURE in 1h 42m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ad4754d8187d5c85787489544dcd8cf2943972e2","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-18 21:37:10.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9d6f5ac85cf74f2cbfe0e292fa1339b0/index.html : SUCCESS in 17m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/118e19f3a39d4ee7b48a350a5b24b6d1/index.html : SUCCESS in 27m 52s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0133c6295f184fb692a847203f00263c/index.html : SUCCESS in 20m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/2/check/real-db-upgrade_nova_percona_user_001:th-percona/97ae4b0bfd504f12afcd6a04b72902b2/index.html : SUCCESS in 23m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/dfa57fbd028c45d6a218da8e2ac008a5/index.html : SUCCESS in 21m 33s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/2/check/real-db-upgrade_nova_percona_user_002:th-percona/79ed002c04224e7eb8794030144068a6/index.html : SUCCESS in 21m 34s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e633f3b24788b2241a3139977cd088cdc594a21","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-06-19 08:51:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e59cc2062669858ffa6cb031b4c92044e8e81296","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-19 09:21:51.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/57cb069d953549ae88ad80d4b40ae64e/index.html : SUCCESS in 21m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c21ac95c0e1c4ad1819125112d514100/index.html : SUCCESS in 26m 44s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a554cdef42d64c6fb68a5ce225f9010d/index.html : SUCCESS in 20m 27s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/3/check/real-db-upgrade_nova_percona_user_001:th-percona/e1163a7847e64d1d9d67c621e8453ce1/index.html : SUCCESS in 29m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/aa6b4562be4143d0b17c3df53646dbfd/index.html : SUCCESS in 17m 58s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/3/check/real-db-upgrade_nova_percona_user_002:th-percona/6c132b8c08b14d9c9507ddba6cda4f62/index.html : SUCCESS in 20m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"e661b58f95e7f10d1c5be8e05a69ef09e4a6b84d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-06-19 09:49:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/3/Nova-ML2-Sriov : SUCCESS in 21m 09s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"6ad7573de921fccb57235719faec9c1eabaca085","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-19 10:01:36.000000000","message":"Patch Set 3: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/3/1718/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"ce8306965fb158da8703f28825be7e5a79da1f95","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-19 10:01:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/3/check/gate-nova-pep8/492a749/ : SUCCESS in 8m 39s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/3/check/gate-nova-docs/1e0a8cf//doc/build/html/ : SUCCESS in 8m 37s\n- gate-nova-python27 http://logs.openstack.org/35/193135/3/check/gate-nova-python27/bfbcba8/ : SUCCESS in 15m 23s\n- check-tempest-dsvm-full http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-full/b280601/ : FAILURE in 1h 03m 04s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-postgres-full/3af2425/ : FAILURE in 1h 00m 15s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-neutron-full/4f8e72b/ : FAILURE in 1h 02m 33s\n- check-grenade-dsvm http://logs.openstack.org/35/193135/3/check/check-grenade-dsvm/dc2f8ac/ : FAILURE in 34m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/3/check/gate-tempest-dsvm-large-ops/0678632/ : SUCCESS in 25m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/3/check/gate-tempest-dsvm-neutron-large-ops/636d07b/ : SUCCESS in 23m 26s\n- check-devstack-dsvm-cells http://logs.openstack.org/35/193135/3/check/check-devstack-dsvm-cells/eb3b2bf/ : FAILURE in 21m 02s\n- gate-nova-python34 http://logs.openstack.org/35/193135/3/check/gate-nova-python34/a166791/ : SUCCESS in 9m 25s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/3/check/gate-nova-tox-functional/3364ac2/ : SUCCESS in 12m 33s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/3/check/check-grenade-dsvm-partial-ncpu/44afbde/ : SUCCESS in 40m 02s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-ironic-pxe_ssh/b7072c5/ : SUCCESS in 31m 32s\n- check-tempest-dsvm-docker http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-docker/20713b2/ : SUCCESS in 36m 45s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-nova-v21-full/fd7078e/ : FAILURE in 1h 04m 20s\n- check-tempest-dsvm-cells http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-cells/49589fd/ : FAILURE in 55m 19s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-full-ceph/e04c781/ : FAILURE in 1h 02m 55s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/3/check/check-tempest-dsvm-multinode-full/e3a4a1e/ : FAILURE in 1h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78ef01c797d34d38b5bfb3ddd0664ff3627d15cd","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-19 10:04:52.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: unrecognized error\n- check-grenade-dsvm: unrecognized error\n- check-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-nova-v21-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"4a60879134b1153d04ccbcb514b97632a54f3fcc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-19 10:28:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/3/check/dsvm-tempest-xen/6a390d3 : FAILURE in 1h 36m 32s","accounts_in_message":[],"_revision_number":3},{"id":"3675bb5fd4dc106b6efce3a40184a2e9024cba2a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-19 10:35:53.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/3/check-ibm-tempest-dsvm-full/d2bc675 : SUCCESS in 47m 58s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/3/check-ibm-tempest-dsvm-postgres-full/4e60b3a : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b586ef2eac2b7e9a3ceacad23201f831e8bb5236","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-06-19 13:12:44.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/193135/3/ : FAILURE in 3h 12m 01s","accounts_in_message":[],"_revision_number":3},{"id":"6c759ffb3c7f48f7f929c5ee95a5cb8adb6fa4f0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-06-22 23:24:24.000000000","message":"Patch Set 3:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/3 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/193135/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6c4fbd178d0a6c7222d5f9f121ebcd847e9c899c","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-06-24 16:07:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"545d2b2dbf312589db57f7c97d74db61b58babe3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-06-24 16:36:23.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/297e9e37f4c044678864831c8ac2bdde/index.html : SUCCESS in 21m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9fde5776ec5d4953a9fd11dcec7442fe/index.html : SUCCESS in 26m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1c3f000d271844bea6fbacaa25e4bff0/index.html : SUCCESS in 18m 09s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/4/check/real-db-upgrade_nova_percona_user_001:th-percona/e3fe3a52abe74e399ae94dca646ac4a8/index.html : SUCCESS in 27m 05s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d17934a488a34597899fd9c943d80d58/index.html : SUCCESS in 18m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/4/check/real-db-upgrade_nova_percona_user_002:th-percona/b0fb2ad0cccd435bb27d8f0bcaefe06e/index.html : SUCCESS in 20m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"705dea5297a2a2ce7708260f9646b0ef52f374dd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-06-24 17:05:29.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/4/check-ibm-tempest-dsvm-full/97c187e : SUCCESS in 56m 36s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/4/check-ibm-tempest-dsvm-postgres-full/5e48336 : SUCCESS in 44m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"45927920ee309997a3b774e166ee44dbe0583c4e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-06-24 17:15:28.000000000","message":"Patch Set 4: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/4/2207/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"7a07a9285cd0c900c5b93c64a86134a9ada1198c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-06-24 17:17:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/4/check/gate-nova-pep8/a7be22c/ : SUCCESS in 6m 27s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/4/check/gate-nova-docs/6f8bf86//doc/build/html/ : SUCCESS in 6m 52s\n- gate-nova-python27 http://logs.openstack.org/35/193135/4/check/gate-nova-python27/f2ba9ef/ : SUCCESS in 14m 19s\n- check-tempest-dsvm-full http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-full/ab88b01/ : FAILURE in 54m 53s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-postgres-full/2216150/ : FAILURE in 52m 11s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-neutron-full/8258eb2/ : FAILURE in 1h 04m 39s\n- check-grenade-dsvm http://logs.openstack.org/35/193135/4/check/check-grenade-dsvm/f5ba36f/ : FAILURE in 36m 48s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/4/check/gate-tempest-dsvm-large-ops/070e43a/ : SUCCESS in 21m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/4/check/gate-tempest-dsvm-neutron-large-ops/7e8df81/ : SUCCESS in 20m 52s\n- check-devstack-dsvm-cells http://logs.openstack.org/35/193135/4/check/check-devstack-dsvm-cells/4bf2474/ : FAILURE in 21m 58s\n- gate-nova-python34 http://logs.openstack.org/35/193135/4/check/gate-nova-python34/4372f9f/ : SUCCESS in 7m 36s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/4/check/gate-nova-tox-functional/d58aa56/ : SUCCESS in 13m 20s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/4/check/check-grenade-dsvm-partial-ncpu/a7d90fa/ : SUCCESS in 37m 08s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-ironic-pxe_ssh/bdb59e3/ : SUCCESS in 34m 38s\n- check-tempest-dsvm-docker http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-docker/f51c722/ : FAILURE in 30m 13s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-nova-v21-full/e1d5828/ : FAILURE in 58m 21s\n- check-tempest-dsvm-cells http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-cells/0682107/ : FAILURE in 42m 59s (non-voting)\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-full-ceph/4a2805c/ : FAILURE in 58m 38s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/4/check/check-tempest-dsvm-multinode-full/55a8a3f/ : FAILURE in 1h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ac5cc5339b76d991813cf3949e6422b63b4ee3eb","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-06-24 17:19:37.000000000","message":"Patch Set 4:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: unrecognized error\n- check-grenade-dsvm: unrecognized error\n- check-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-nova-v21-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"2c18393ab407e9f210db38aec9cfbc95ff411529","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-06-24 18:00:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/4/check/dsvm-tempest-xen/269a746 : FAILURE in 1h 51m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2365ca6dafe112c7ddd65b464ec1e163dc5aba5d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-03 06:23:46.000000000","message":"Patch Set 4:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/4 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/193135/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f91435e8f6e6110676eb4bb03286025bbf0cc8c7","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-07-03 12:02:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28c3307386ac3d6636a7af8188f0e4416a0fc543","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-03 12:17:31.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/5/check-ibm-tempest-dsvm-full/3e40bc4 : FAILURE in 14m 27s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/5/check-ibm-tempest-dsvm-postgres-full/909a210 : FAILURE in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fb93f9d1ccc6f6cc50f5cc4a03430a09ac089c88","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-03 13:10:16.000000000","message":"Patch Set 5: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/5/3272/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"b0101909de52a3f46eb0e6918d3971ab2282e3de","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-03 13:37:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/5/check/gate-nova-pep8/8a8d37d/ : SUCCESS in 9m 12s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/5/check/gate-nova-docs/faa5a36//doc/build/html/ : SUCCESS in 10m 25s\n- gate-nova-python27 http://logs.openstack.org/35/193135/5/check/gate-nova-python27/7a1daba/ : SUCCESS in 13m 54s\n- gate-nova-python34 http://logs.openstack.org/35/193135/5/check/gate-nova-python34/b80ca9e/ : SUCCESS in 11m 56s\n- check-tempest-dsvm-full http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-full/3f075ac/ : FAILURE in 1h 02m 31s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-postgres-full/1d1337f/ : FAILURE in 1h 03m 20s\n- check-tempest-dsvm-neutron http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-neutron/dab34e0/ : FAILURE in 43m 02s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-neutron-full/401bc71/ : FAILURE in 1h 13m 49s\n- check-grenade-dsvm http://logs.openstack.org/35/193135/5/check/check-grenade-dsvm/e9725de/ : FAILURE in 43m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/5/check/gate-tempest-dsvm-large-ops/f9d4761/ : SUCCESS in 34m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/5/check/gate-tempest-dsvm-neutron-large-ops/9616183/ : SUCCESS in 30m 47s\n- check-devstack-dsvm-cells http://logs.openstack.org/35/193135/5/check/check-devstack-dsvm-cells/27184fc/ : FAILURE in 20m 29s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/5/check/gate-nova-tox-functional/b61bbce/ : SUCCESS in 9m 55s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/5/check/check-grenade-dsvm-partial-ncpu/2aca017/ : SUCCESS in 50m 06s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-ironic-pxe_ssh/7218fb9/ : SUCCESS in 36m 20s\n- check-tempest-dsvm-docker http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-docker/a189c42/ : SUCCESS in 41m 58s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-nova-v21-full/d6eba08/ : FAILURE in 54m 14s\n- check-tempest-dsvm-cells http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-cells/2604f1e/ : FAILURE in 1h 00m 31s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-full-ceph/03a36d4/ : FAILURE in 1h 04m 33s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-multinode-full/d0b7c69/ : FAILURE in 1h 30m 24s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-multinode-smoke/61c777f/ : FAILURE in 21m 34s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/35/193135/5/check/check-tempest-dsvm-neutron-multinode-smoke/0774c26/ : FAILURE in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff550a3601226de71d02182a43982a8b3c84d77e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-03 13:37:51.000000000","message":"Patch Set 5:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: unrecognized error\n- check-grenade-dsvm: unrecognized error\n- check-tempest-dsvm-cells: unrecognized error\n- check-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron: unrecognized error\n- check-tempest-dsvm-nova-v21-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":5},{"id":"2e20acbd46baeb14c2651d1dcd0540eb7970b0c9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-03 13:52:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/5/check/dsvm-tempest-xen/f6e8aea : FAILURE in 1h 49m 07s","accounts_in_message":[],"_revision_number":5},{"id":"f2b22c32f4935ece9446a9890677f8bb3d567dc0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-03 14:53:19.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/193135/5/ : FAILURE in 2h 09m 56s","accounts_in_message":[],"_revision_number":5},{"id":"9c731548c7624266d8878c83bb6d0a69b0b3d253","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-05 13:02:34.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/939eda80f2194857bafdc4141328cf31/index.html : SUCCESS in 17m 16s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6c45a6f008c44ca2bae7e04c5e578126/index.html : SUCCESS in 24m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fa59f67142b1453f96023a5cd023bf10/index.html : SUCCESS in 16m 39s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/5/check/real-db-upgrade_nova_percona_user_001:th-percona/02d38dc0cc514bc2b28c1f0dd7a0fa86/index.html : SUCCESS in 20m 23s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6cd1a80e3b0b465c8e22efdb52f9921c/index.html : SUCCESS in 17m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/5/check/real-db-upgrade_nova_percona_user_002:th-percona/f521dde5018b40ca81d3f530fbf91452/index.html : SUCCESS in 15m 41s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"46f01a9f9a4fb47633c91dbd020c19ca0a1d1464","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-07-07 10:11:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"22db2e8072c852ef46ec50965528254b69b33eef","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-07 10:41:48.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/48c6b906638c47ed95e7db48a332d227/index.html : SUCCESS in 20m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/52dd74b03a034bf580e8334aff40926f/index.html : SUCCESS in 29m 06s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/82fa7541b99243118e6c8240277d8a75/index.html : SUCCESS in 17m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/6/check/real-db-upgrade_nova_percona_user_001:th-percona/5c7de1ccc72d4b0fade5ed50da444068/index.html : SUCCESS in 24m 52s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c35b209057b44b5956677733ab65c59/index.html : SUCCESS in 17m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/6/check/real-db-upgrade_nova_percona_user_002:th-percona/c47c4eab20a94d7fba7d51987cb4a040/index.html : SUCCESS in 18m 01s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"b567c31aec8e4189a52f6c1b07b39d83da2960f7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-07 11:36:27.000000000","message":"Patch Set 6: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/6/3571/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"595fbe2ca9eb02f756808d6d61d8e951cf0e635b","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-07-07 11:43:41.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"004e8a6b6729f6ac08e7ff00f08dbf0049bcd5a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-07 12:02:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/6/check/gate-nova-pep8/e5bbdf0/ : SUCCESS in 9m 08s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/6/check/gate-nova-docs/9674024//doc/build/html/ : SUCCESS in 9m 18s\n- gate-nova-python27 http://logs.openstack.org/35/193135/6/check/gate-nova-python27/6857dbc/ : SUCCESS in 14m 07s\n- gate-nova-python34 http://logs.openstack.org/35/193135/6/check/gate-nova-python34/c238460/ : SUCCESS in 11m 55s\n- check-tempest-dsvm-full http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-full/7cd64ae/ : FAILURE in 1h 22m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-postgres-full/1ac312b/ : FAILURE in 1h 08m 58s\n- check-tempest-dsvm-neutron http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-neutron/bda49d7/ : FAILURE in 46m 11s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-neutron-full/8f27024/ : FAILURE in 1h 32m 35s\n- check-grenade-dsvm http://logs.openstack.org/35/193135/6/check/check-grenade-dsvm/9383bc3/ : FAILURE in 40m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/6/check/gate-tempest-dsvm-large-ops/bd334dd/ : SUCCESS in 35m 03s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/6/check/gate-tempest-dsvm-neutron-large-ops/69bd4c6/ : SUCCESS in 33m 34s\n- check-devstack-dsvm-cells http://logs.openstack.org/35/193135/6/check/check-devstack-dsvm-cells/1d17361/ : FAILURE in 28m 36s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/6/check/gate-nova-tox-functional/0541508/ : SUCCESS in 13m 25s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/6/check/check-grenade-dsvm-partial-ncpu/a35f4cc/ : SUCCESS in 51m 04s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-ironic-pxe_ssh/f9ed391/ : SUCCESS in 44m 36s\n- check-tempest-dsvm-docker http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-docker/003d998/ : FAILURE in 44m 58s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-nova-v21-full/2b98d50/ : FAILURE in 1h 16m 41s\n- check-tempest-dsvm-cells http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-cells/c3cd339/ : FAILURE in 57m 35s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-full-ceph/bff33bb/ : FAILURE in 1h 14m 39s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-multinode-full/0820df1/ : FAILURE in 1h 47m 10s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-multinode-smoke/d875716/ : FAILURE in 57m 44s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/35/193135/6/check/check-tempest-dsvm-neutron-multinode-smoke/b59a273/ : FAILURE in 1h 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"73167847cd287768a5871eb6e11f7f26b387f0f9","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-07 12:03:11.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-devstack-dsvm-cells: unrecognized error\n- check-grenade-dsvm: unrecognized error\n- check-tempest-dsvm-cells: unrecognized error\n- check-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1286818 https://bugs.launchpad.net/bugs/1311066 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1311066 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-neutron: https://bugs.launchpad.net/bugs/1311066\n- check-tempest-dsvm-nova-v21-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1311066 https://bugs.launchpad.net/bugs/1352498\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1421895 https://bugs.launchpad.net/bugs/1352498\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"5ff35b9a6ab13767b1846cea528c95c0af1ee33f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-07 12:19:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/6/check/dsvm-tempest-xen/6c7abb8 : FAILURE in 2h 06m 54s","accounts_in_message":[],"_revision_number":6},{"id":"dfeedf502008f65ae1f8c93efb9936142040aacf","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-07 12:35:24.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/193135/6/ : FAILURE in 2h 22m 29s","accounts_in_message":[],"_revision_number":6},{"id":"bae88e04f3f95b83e769439fc62cbc3cc944167a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-07 13:16:14.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/6/check-ibm-tempest-dsvm-full/9fe4fff : SUCCESS in 50m 44s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/6/check-ibm-tempest-dsvm-postgres-full/c0f43c9 : SUCCESS in 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3d81b9b102ee945850649b1e9e21c3fcc973ba60","author":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"date":"2015-07-07 14:13:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e9163eac72c5b7a5cde70b5d254ca4530dfe9f2e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-07 14:43:35.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b78a4facc3ce43868aa68561253c567f/index.html : SUCCESS in 21m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c784c29759a44f3680c59a0e65924ef7/index.html : SUCCESS in 29m 05s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b856dda6eaa1439bbee36484dfb3906f/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/7/check/real-db-upgrade_nova_percona_user_001:th-percona/eb1b75c920064cfda55a436e1d7da4a9/index.html : SUCCESS in 20m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/b27d7af9d6854d4fb9d3567cf67b5b97/index.html : SUCCESS in 18m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/7/check/real-db-upgrade_nova_percona_user_002:th-percona/ac6659ce5bb84ce7891a9ee1ec3d7c15/index.html : SUCCESS in 18m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"69b27fe6070974f4f71cd136bb217a1fc3ea2202","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-07 15:20:33.000000000","message":"Patch Set 7: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/7/3602/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":7},{"id":"82d392a88916de126e95108a8f751849015307eb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-07 15:21:46.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/7/check-ibm-tempest-dsvm-full/69a17fc : SUCCESS in 52m 46s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/7/check-ibm-tempest-dsvm-postgres-full/9b6217d : SUCCESS in 1h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a4fc13ed78500375cf15a34bb2a64923c090ffca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-07 15:52:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/7/check/gate-nova-pep8/9cb3c28/ : SUCCESS in 10m 07s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/7/check/gate-nova-docs/cdd3cc7//doc/build/html/ : SUCCESS in 8m 48s\n- gate-nova-python27 http://logs.openstack.org/35/193135/7/check/gate-nova-python27/e3d8800/ : FAILURE in 14m 24s\n- gate-nova-python34 http://logs.openstack.org/35/193135/7/check/gate-nova-python34/4644048/ : SUCCESS in 12m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-full/636e4b8/ : FAILURE in 1h 17m 55s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-postgres-full/02db50b/ : FAILURE in 1h 10m 26s\n- check-tempest-dsvm-neutron http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-neutron/e4a6700/ : FAILURE in 2m 00s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-neutron-full/e475a14/ : FAILURE in 1m 45s\n- check-grenade-dsvm http://logs.openstack.org/35/193135/7/check/check-grenade-dsvm/034ef8f/ : FAILURE in 49m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/7/check/gate-tempest-dsvm-large-ops/0bcff69/ : SUCCESS in 34m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/7/check/gate-tempest-dsvm-neutron-large-ops/08e206d/ : SUCCESS in 36m 19s\n- check-devstack-dsvm-cells http://logs.openstack.org/35/193135/7/check/check-devstack-dsvm-cells/654f3df/ : FAILURE in 28m 57s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/7/check/gate-nova-tox-functional/4929dc5/ : SUCCESS in 13m 55s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/7/check/check-grenade-dsvm-partial-ncpu/7549eae/ : SUCCESS in 49m 14s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-ironic-pxe_ssh/10ef19f/ : FAILURE in 1m 43s\n- check-tempest-dsvm-docker http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-docker/9cbca0a/ : FAILURE in 49m 14s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-nova-v21-full/9deb2f6/ : FAILURE in 1h 16m 44s\n- check-tempest-dsvm-cells http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-cells/6648ade/ : FAILURE in 2m 07s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-full-ceph/11f3354/ : FAILURE in 2m 02s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-multinode-full/a27babe/ : FAILURE in 24m 18s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-multinode-smoke/9033a06/ : FAILURE in 38m 22s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/35/193135/7/check/check-tempest-dsvm-neutron-multinode-smoke/848af86/ : FAILURE in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bcc47518d550254bd498ee61e49b52d57528a8f2","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-07 16:08:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/7/check/dsvm-tempest-xen/e2109f1 : FAILURE in 1h 54m 47s","accounts_in_message":[],"_revision_number":7},{"id":"087f73dcb90af3c4cae489049b9b8d99be4d1b53","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-07 17:09:36.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/193135/7/ : FAILURE in 1h 49m 41s","accounts_in_message":[],"_revision_number":7},{"id":"07ecea0789478bff3ac0a5d720bbb82c7bb482e1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-07 19:26:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/7/Nova-ML2-Sriov : SUCCESS in 55m 31s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"053fbbd9bc7e75e29a627e092063bc106f9c57d5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-13 20:27:09.000000000","message":"Patch Set 7:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/7 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/193135/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"e4b8e4b392bb5d0a8e0f37aad927d57e09d99f7d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-07-28 16:42:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)\n\nAlso needs unit tests.","accounts_in_message":[],"_revision_number":7},{"id":"8980cc3c969dadfff4e2fe4575ab810b59aa51e4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-08-04 11:08:12.000000000","message":"Patch Set 7: Code-Review-2\n\nSorry, we have now hit the non-priority feature freeze for Liberty. You will need to resubmit this blueprint for Mitaka or apply for an exception. For more details on why this is happening, and the rest of the process details, please see: https://wiki.openstack.org/wiki/Nova/Liberty_Release_Schedule","accounts_in_message":[],"_revision_number":7},{"id":"1199f3ff5b9e0ee837d28d1f4db24e09c6659340","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2015-08-27 13:20:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5b63272140d192b9730a3e3f0fb1c1bf22437dd7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-27 13:29:43.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/193135/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"2683bea52fe611b6fdccb187ebad636f8f06b78a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-27 13:53:04.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f27974a68e5a4ad1bf9164fa52bfd294/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/43fe88c4e7854a6685111d7049159bc6/index.html : SUCCESS in 14m 55s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/06689dc1eb5d4d78bf8823892abe4b6b/index.html : SUCCESS in 14m 38s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/8/check/real-db-upgrade_nova_percona_user_001:th-percona/6ad957578be84dd0acd43c38b22395a3/index.html : SUCCESS in 17m 44s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5e1994e347b44461a9f7cd5af95d32d7/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/8/check/real-db-upgrade_nova_percona_user_002:th-percona/0a3d4960ad2540128eb73a996b423ed3/index.html : SUCCESS in 12m 20s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"27946615eab5790ab66218fc9d7a2d5f82923039","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-27 14:25:32.000000000","message":"Patch Set 8: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/8/10032/index.html : FAILURE in 58 m 11 s","accounts_in_message":[],"_revision_number":8},{"id":"7dcfbdc5f70f1a8983a8d402df8d5f64c777f1f9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-27 14:43:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/8/check/dsvm-tempest-xen/a7bc354 : SUCCESS in 1h 21m 39s","accounts_in_message":[],"_revision_number":8},{"id":"37ef8e55388c051544925361cfbec784878f17fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-27 15:04:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/8/check/gate-nova-pep8/e36d350/ : SUCCESS in 9m 09s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/8/check/gate-nova-docs/2e24739//doc/build/html/ : SUCCESS in 33m 36s\n- gate-nova-python27 http://logs.openstack.org/35/193135/8/check/gate-nova-python27/bac48aa/ : SUCCESS in 17m 21s\n- gate-nova-python34 http://logs.openstack.org/35/193135/8/check/gate-nova-python34/af85057/ : SUCCESS in 15m 08s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-full/c7015fd/ : FAILURE in 47m 30s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-postgres-full/5add1e0/ : FAILURE in 39m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-neutron-full/7d38204/ : FAILURE in 1h 02m 15s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/8/check/gate-grenade-dsvm/273e8f9/ : SUCCESS in 39m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-large-ops/7adb472/ : SUCCESS in 28m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-neutron-large-ops/abf73f2/ : SUCCESS in 27m 31s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/8/check/gate-nova-tox-functional/5f447c9/ : SUCCESS in 18m 21s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/35/193135/8/check/gate-grenade-dsvm-partial-ncpu/2a08071/ : SUCCESS in 39m 12s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-ironic-pxe_ssh/7f32f77/ : SUCCESS in 39m 48s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-nova-v21-full/9be3c79/ : FAILURE in 45m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-cells/54df19b/ : FAILURE in 41m 37s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-full-ceph/bfe1a56/ : FAILURE in 47m 29s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/8/check/gate-tempest-dsvm-multinode-full/e03313b/ : FAILURE in 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"77dcc9bedca14a81375b853936ccb99907f66be5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-27 16:18:06.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/8/check-ibm-tempest-dsvm-full/85a5701 : SUCCESS in 1h 16m 16s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/193135/8/check-ibm-tempest-dsvm-postgres-full/327e0f4 : SUCCESS in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8335827a27b8471600f003a3e5b2e107f3e41620","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2015-09-03 18:34:50.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"137b3ec15525ecc347e7ec52c8ed36a8b2dfa6a7","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-11-29 18:11:41.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ea4fb0f42deddbe8ced4ee5a3913d0d680c08cdf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-29 18:28:29.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/9/Nova-ML2-Sriov : FAILURE in 14m 59s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"ea611d86d9f8913e069c101c7a396ec4332aef35","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-29 18:30:01.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5ef4d992cff54306beae262774f0c2aa/index.html : SUCCESS in 12m 26s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a321951633c14814bcc3feb7d24a93bc/index.html : SUCCESS in 16m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d1a5d3929f3e40beaf540b31cab2b42e/index.html : SUCCESS in 13m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/9/check/real-db-upgrade_nova_percona_user_001:th-percona/f57dba160bd74f50ad242f73243b3fc7/index.html : SUCCESS in 16m 58s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6fbe7a743dbe4600ad258db54d469a2d/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/9/check/real-db-upgrade_nova_percona_user_002:th-percona/1f73767f40384c78a96934d8a69882ae/index.html : SUCCESS in 14m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"eb835551f2b803451f2ba4e031df26e4758e9a78","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-29 18:37:47.000000000","message":"Patch Set 9:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/9 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/193135/9 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"3583e1e587f06aeb3545ef80cc0ba02a29ea65dd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-29 18:38:35.000000000","message":"Patch Set 9: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/9/17366/index.html : FAILURE in 23 m 41 s","accounts_in_message":[],"_revision_number":9},{"id":"1eab01d8305f22ba7437e96d3e7366da3a954c91","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-11-29 19:14:08.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"acf92272c37bec5ab740b682717110ecf8775b4e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-29 19:15:09.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/193135/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"210b72e984904912b5dd85eb6c3cf8190d829efa","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-29 19:31:42.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2e12ad4151784147b80a9c8aca3218ce/index.html : SUCCESS in 12m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3d74316019ef401eb288c5e67fff6b9d/index.html : SUCCESS in 16m 06s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/cf429e48834e49cb95cbb976c6d713fe/index.html : SUCCESS in 15m 38s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/10/check/real-db-upgrade_nova_percona_user_001:th-percona/9fde73a302f04bb7897d138c3d762371/index.html : SUCCESS in 16m 09s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a8dba9c047664945adb2b5143a2be11d/index.html : SUCCESS in 14m 53s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/10/check/real-db-upgrade_nova_percona_user_002:th-percona/62df2281176645feb3b99bb466ba0134/index.html : SUCCESS in 14m 42s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"1db9d928a823704cb2d8757c18e80572eb964249","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-29 20:07:02.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/10/ : FAILURE in 51m 40s","accounts_in_message":[],"_revision_number":10},{"id":"d2db541e9803aeab6a1e3708283ba818c0df5cce","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-29 20:11:03.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/10/Nova-ML2-Sriov : SUCCESS in 55m 17s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"281b73c597c613afa74829a630a9e9e91f2bf8d3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-29 20:17:07.000000000","message":"Patch Set 10: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/10/17371/index.html : FAILURE in 01 h 00 m 24 s","accounts_in_message":[],"_revision_number":10},{"id":"405d18909689cda749a82e0bb70a79c68a2c320f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-29 20:38:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/10/check/gate-nova-pep8/4225d9c/ : FAILURE in 7m 26s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/10/check/gate-nova-docs/1c76718//doc/build/html/ : SUCCESS in 9m 30s\n- gate-nova-python27 http://logs.openstack.org/35/193135/10/check/gate-nova-python27/e2f984d/ : FAILURE in 11m 11s\n- gate-nova-python34 http://logs.openstack.org/35/193135/10/check/gate-nova-python34/f4f79f6/ : FAILURE in 13m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-full/5f3c9a9/ : FAILURE in 55m 04s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-postgres-full/dfd4956/ : FAILURE in 53m 14s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-neutron-full/172c29c/ : FAILURE in 1h 20m 58s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/10/check/gate-grenade-dsvm/3609c6d/ : SUCCESS in 47m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-large-ops/e4ca1ef/ : SUCCESS in 36m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-neutron-large-ops/4256171/ : SUCCESS in 34m 29s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/10/check/gate-nova-tox-functional/437004d/ : SUCCESS in 13m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/10/check/gate-grenade-dsvm-multinode/84809f6/ : SUCCESS in 47m 11s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0f2fbc5/ : SUCCESS in 46m 19s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-nova-v20-api/d54556b/ : FAILURE in 34m 18s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-nova-v20-api-legacy/0305fdf/ : FAILURE in 44m 32s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-cells/92d5d7e/ : FAILURE in 47m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-full-ceph/23dd8d4/ : FAILURE in 47m 33s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/10/check/gate-tempest-dsvm-multinode-full/481867d/ : FAILURE in 1h 05m 21s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/10/check/gate-compute-api-guide/ee34fb8//api-guide/build/html/ : SUCCESS in 8m 47s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/10/check/gate-nova-python27-constraints/0a6c196/ : FAILURE in 11m 37s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/10/check/gate-nova-python34-constraints/767244e/ : FAILURE in 11m 10s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/10/check/gate-nova-docs-constraints/48584ca//doc/build/html/ : SUCCESS in 10m 12s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/10/check/gate-nova-pep8-constraints/11869ab/ : FAILURE in 6m 42s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/10/check/gate-nova-tox-constraints-functional-constraints/57159fa/ : SUCCESS in 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bde3e4a3f7b9cd0a17c0581d4e3fb3a44e588e88","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-29 20:42:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/10/check/dsvm-tempest-xen/bbddf28 : SUCCESS in 1h 27m 56s","accounts_in_message":[],"_revision_number":10},{"id":"8263f744652dd6ec89fedae41b24088a97488dc2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-29 20:44:47.000000000","message":"Patch Set 10:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/10 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/193135/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"3a9b5b45b120375c36401965f4505f675ef312f0","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-11-29 21:08:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"31d0a07a899f59920c19ad79bb5519a46cd7dd07","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-29 21:25:33.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/cba053a785634e06b5ebd68277af2749/index.html : SUCCESS in 15m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3ef0e08de30c4c539d2e38a68dfd8a1d/index.html : SUCCESS in 15m 33s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f6c7d022e3304722be9a47b60b241d4e/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/11/check/real-db-upgrade_nova_percona_user_001:th-percona/ef71b9d91f0d45debf1838fe5cf64619/index.html : SUCCESS in 16m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2d99208328134ed68c7f0f1863703efb/index.html : SUCCESS in 12m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/11/check/real-db-upgrade_nova_percona_user_002:th-percona/4884b813bb36414da37807636fe69169/index.html : SUCCESS in 12m 40s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"115983969d1eafde85bbd347e1cbf8b062f72e57","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-29 22:10:32.000000000","message":"Patch Set 11: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/11/17375/index.html : FAILURE in 59 m 32 s","accounts_in_message":[],"_revision_number":11},{"id":"28e5fcf53466b541afd37d254f733a23b7395a77","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-29 22:14:26.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/11/ : FAILURE in 1h 04m 39s","accounts_in_message":[],"_revision_number":11},{"id":"9fe2e344214df79593291e94762fbc87a4cd55e3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-29 22:15:52.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/11/Nova-ML2-Sriov : SUCCESS in 1h 05m 53s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"14fe13f05089bd121c2847f448b53ba1b39358c4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-29 22:24:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/11/check/gate-nova-pep8/0824568/ : FAILURE in 11m 06s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/11/check/gate-nova-docs/3e6fa53//doc/build/html/ : SUCCESS in 14m 51s\n- gate-nova-python27 http://logs.openstack.org/35/193135/11/check/gate-nova-python27/311db9f/ : FAILURE in 14m 49s\n- gate-nova-python34 http://logs.openstack.org/35/193135/11/check/gate-nova-python34/de063ae/ : SUCCESS in 13m 15s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-full/7070ad0/ : FAILURE in 44m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-postgres-full/2b5de66/ : FAILURE in 44m 26s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-neutron-full/132b67c/ : FAILURE in 1h 04m 00s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/11/check/gate-grenade-dsvm/7585a5b/ : SUCCESS in 40m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-large-ops/ca248ee/ : SUCCESS in 27m 57s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-neutron-large-ops/6da8f37/ : SUCCESS in 28m 36s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/11/check/gate-nova-tox-functional/b95a892/ : SUCCESS in 16m 36s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/11/check/gate-grenade-dsvm-multinode/7e89c25/ : SUCCESS in 48m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/36b3b7d/ : SUCCESS in 51m 52s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-nova-v20-api/cb15834/ : FAILURE in 40m 47s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-nova-v20-api-legacy/d333122/ : FAILURE in 41m 36s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-cells/1daae6e/ : FAILURE in 45m 22s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-full-ceph/764ba8a/ : FAILURE in 54m 35s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/11/check/gate-tempest-dsvm-multinode-full/43d44fb/ : FAILURE in 1h 13m 15s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/11/check/gate-compute-api-guide/81cfb73//api-guide/build/html/ : SUCCESS in 8m 25s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/11/check/gate-nova-python27-constraints/974d8c3/ : FAILURE in 14m 34s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/11/check/gate-nova-python34-constraints/5ca9ff7/ : SUCCESS in 12m 43s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/11/check/gate-nova-docs-constraints/89a2c36//doc/build/html/ : SUCCESS in 14m 52s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/11/check/gate-nova-pep8-constraints/eddcce5/ : FAILURE in 9m 44s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/11/check/gate-nova-tox-constraints-functional-constraints/958aa57/ : SUCCESS in 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0b8761ebe33df86d1b88cc9d00ff3f0b924190fd","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-29 22:28:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/11/check/dsvm-tempest-xen/98f8da0 : FAILURE in 1h 19m 05s","accounts_in_message":[],"_revision_number":11},{"id":"0a20978b4b336d5fb207a64b32e37644f170a845","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-29 23:15:09.000000000","message":"Patch Set 11:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/11 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/193135/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"e7bf7e09c00df4f350c8e05f4a64461178ecaedc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-11-30 04:01:05.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/11/check/check-ibm-tempest-dsvm-full/23433b8/ : SUCCESS in 1h 00m 30s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/11/check/check-ibm-tempest-dsvm-postgres-full/6ded3fe/ : SUCCESS in 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"851a9fae7cbcbda160e095e8585eedb0911eaaa9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-12-04 09:42:55.000000000","message":"Patch Set 11: -Code-Review\n\nremoving -2 now the blueprint is approved.","accounts_in_message":[],"_revision_number":11},{"id":"98aa49d3cdbc84ccef58d0f7d0434175fd7f69a1","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-04 21:38:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a4c6ccc77a372184a7ff53f0ca7a5de8c951a25c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-04 21:54:22.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/193135/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"a975568e4ede630d1492c73df129422e147ea007","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-04 21:56:36.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e98403091954423f860fedf75f339fb8/index.html : SUCCESS in 14m 01s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c02712a7dcc145ccb36cfb9b01a86b76/index.html : SUCCESS in 14m 05s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c60d824cd1024ae4b7d04072885a77cd/index.html : SUCCESS in 14m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/12/check/real-db-upgrade_nova_percona_user_001:th-percona/093d174cbd614c53b7473a0e0dade02e/index.html : SUCCESS in 17m 04s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e5849f1eba5743bc9b21d382298a5e24/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/12/check/real-db-upgrade_nova_percona_user_002:th-percona/29db308183174f47893f20951dfd66e9/index.html : SUCCESS in 12m 30s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"8a43bede0d3d703744aae9602bfc52d2d91ea6d3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-04 22:36:02.000000000","message":"Patch Set 12: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/12/18177/index.html : FAILURE in 54 m 46 s","accounts_in_message":[],"_revision_number":12},{"id":"53cec428ab37c04bf9b1101eaed89ccfcd17d095","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-04 22:56:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/12/check/dsvm-tempest-xen/fb209e8 : FAILURE in 1h 16m 06s","accounts_in_message":[],"_revision_number":12},{"id":"3527810a6cc19a097372ad51c76e863ad5ca8a42","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-04 23:10:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/12/check/gate-nova-pep8/0d369c8/ : SUCCESS in 8m 50s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/12/check/gate-nova-docs/bd2fdab//doc/build/html/ : SUCCESS in 14m 25s\n- gate-nova-python27 http://logs.openstack.org/35/193135/12/check/gate-nova-python27/ec116fc/ : FAILURE in 14m 51s\n- gate-nova-python34 http://logs.openstack.org/35/193135/12/check/gate-nova-python34/acfbb03/ : SUCCESS in 12m 37s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-full/ea90f77/ : FAILURE in 47m 26s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-postgres-full/c165f02/ : FAILURE in 58m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-neutron-full/da8fb16/ : FAILURE in 1h 15m 48s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/12/check/gate-grenade-dsvm/d39a28c/ : SUCCESS in 44m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-large-ops/5659745/ : SUCCESS in 44m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-neutron-large-ops/0b41ae8/ : SUCCESS in 37m 29s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/12/check/gate-nova-tox-functional/58b8ee5/ : SUCCESS in 17m 43s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/12/check/gate-grenade-dsvm-multinode/30d15c9/ : FAILURE in 54m 43s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/08ecdef/ : SUCCESS in 49m 33s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-nova-v20-api/76ca5a6/ : FAILURE in 41m 08s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-nova-v20-api-legacy/cd4dbd6/ : FAILURE in 42m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-cells/2499724/ : FAILURE in 42m 41s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-full-ceph/f95ef0f/ : FAILURE in 57m 35s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/12/check/gate-tempest-dsvm-multinode-full/d7b6c56/ : FAILURE in 1h 25m 53s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/12/check/gate-compute-api-guide/f6e6526//api-guide/build/html/ : SUCCESS in 9m 16s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/12/check/gate-nova-python27-constraints/2d1ff82/ : FAILURE in 14m 23s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/12/check/gate-nova-python34-constraints/25f5ad6/ : SUCCESS in 13m 21s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/12/check/gate-nova-docs-constraints/a7bcdd6//doc/build/html/ : SUCCESS in 14m 21s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/12/check/gate-nova-pep8-constraints/15c53c8/ : SUCCESS in 10m 29s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/12/check/gate-nova-tox-constraints-functional-constraints/0006d87/ : SUCCESS in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"121f09632d80e53a42624a353ffa95f5c68ed15e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-04 23:28:31.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/12/ : FAILURE in 1h 48m 39s","accounts_in_message":[],"_revision_number":12},{"id":"ea46b1a61c0e0ae2954186c8a597e5015c778868","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-05 11:09:45.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/12/Nova-ML2-Sriov : SUCCESS in 55m 20s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"51d9ab159eab8c3b73c4c5fedbc57acc70f86394","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-05 21:23:00.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a60badaed552b973f9140404766b4b3dfa807dc4","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-05 21:23:56.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5ed3bc2a5ef3efc1b05c6fced9be9ec19e668a41","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-05 21:24:38.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/193135/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"cb86ec1351781d63b0e7ed3a06bfddfc72ae8986","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-05 21:35:55.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/193135/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"56db0f87f0905351e6fc97a692237ddb8626b54a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-05 21:48:34.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/113643e5cb9f4598b54276cd237a62dd/index.html : SUCCESS in 15m 07s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5fbe2f5de44a4fe593322b32bddc7ed4/index.html : SUCCESS in 16m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7c05dffdb302478b940c251e2933feab/index.html : SUCCESS in 12m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/14/check/real-db-upgrade_nova_percona_user_001:th-percona/d1887451cf9b4782b7b122fcbc707462/index.html : SUCCESS in 16m 35s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/14/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5df3078a9bb549b0a6da883c6110be3d/index.html : SUCCESS in 12m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/14/check/real-db-upgrade_nova_percona_user_002:th-percona/288f4c7b77c04a638d3d25c56ca0afae/index.html : SUCCESS in 15m 39s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"e1d648a9ba298bf421968979c3cdbce68a89d576","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-05 22:22:19.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/14/Nova-ML2-Sriov : SUCCESS in 53m 31s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":14},{"id":"67842b3b58346657bce275b7915299b60e14a475","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-05 22:27:36.000000000","message":"Patch Set 14: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/14/18210/index.html : FAILURE in 52 m 11 s","accounts_in_message":[],"_revision_number":14},{"id":"4e7f58d703d4d3707e988b90c9d4929639499cb9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-05 22:41:31.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/14/check/dsvm-tempest-xen/d646a16 : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":14},{"id":"69b44f17490a8d76e3a3fc45cc0268e95155814e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-05 22:49:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/14/check/gate-nova-pep8/3491b00/ : SUCCESS in 7m 43s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/14/check/gate-nova-docs/8190bc0//doc/build/html/ : SUCCESS in 15m 00s\n- gate-nova-python27 http://logs.openstack.org/35/193135/14/check/gate-nova-python27/093e3bc/ : FAILURE in 14m 13s\n- gate-nova-python34 http://logs.openstack.org/35/193135/14/check/gate-nova-python34/e43fafb/ : SUCCESS in 13m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-full/64b8015/ : FAILURE in 46m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-postgres-full/3f7d33c/ : FAILURE in 55m 38s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-neutron-full/59abdd9/ : FAILURE in 1h 02m 18s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/14/check/gate-grenade-dsvm/66d6af7/ : SUCCESS in 41m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-large-ops/16d2e2f/ : SUCCESS in 31m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-neutron-large-ops/1cd7709/ : SUCCESS in 34m 02s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/14/check/gate-nova-tox-functional/54ed78c/ : SUCCESS in 17m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/14/check/gate-grenade-dsvm-multinode/fc3bfbd/ : FAILURE in 44m 26s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/746c402/ : SUCCESS in 50m 20s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-nova-v20-api/2998e14/ : FAILURE in 44m 35s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-nova-v20-api-legacy/809fe95/ : FAILURE in 41m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-cells/3d130a7/ : FAILURE in 42m 34s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-full-ceph/eeaa90d/ : FAILURE in 52m 14s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/14/check/gate-tempest-dsvm-multinode-full/e0284fb/ : FAILURE in 1h 22m 26s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/14/check/gate-compute-api-guide/b0cfeda//api-guide/build/html/ : SUCCESS in 8m 36s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/14/check/gate-nova-python27-constraints/05c463a/ : FAILURE in 14m 42s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/14/check/gate-nova-python34-constraints/a626542/ : SUCCESS in 15m 05s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/14/check/gate-nova-docs-constraints/498a895//doc/build/html/ : SUCCESS in 12m 18s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/14/check/gate-nova-pep8-constraints/d98f4e2/ : SUCCESS in 9m 43s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/14/check/gate-nova-tox-constraints-functional-constraints/ad05328/ : SUCCESS in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a20fa6c2e2fa8571eb5a56bb88ce01d3eedd529b","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-06 11:08:48.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"feb81d6dd5e12317aa8f6e00c35345e3b3cfb6ab","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-06 11:21:42.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/193135/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"3f40781233f04e504b31144e9e6ed381771f8073","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-06 11:28:50.000000000","message":"Patch Set 15: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/15/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/c6e90feff80c4ee5937dd4e01565228f/index.html : SUCCESS in 15m 21s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/15/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4f1908bbc50d46bd9d6cf2f825779b78/index.html : SUCCESS in 16m 29s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/15/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d5422b131d9c425392e7d42a301782c0/index.html : SUCCESS in 15m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/15/check/real-db-upgrade_nova_percona_user_001:th-percona/daf0dfb3b0f643a99af839ae6b1a171c/index.html : SUCCESS in 15m 51s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/15/check/real-db-upgrade_nova_mysql_user_002:th-mysql/70d8e87d9b1a4a30a8eafd6276e53f4a/index.html : SUCCESS in 12m 59s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/15/check/real-db-upgrade_nova_percona_user_002:th-percona/a381922705d146f58d6dc16c6df43d47/index.html : SUCCESS in 15m 14s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":15},{"id":"53eed10d26024051a7472a34b6e5509888293e04","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-06 12:10:21.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/15/Nova-ML2-Sriov : SUCCESS in 54m 08s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"d191e12f2b636f2b0f46a3e0abe8344696684f0c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-06 12:13:28.000000000","message":"Patch Set 15: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/15/18223/index.html : FAILURE in 56 m 51 s","accounts_in_message":[],"_revision_number":15},{"id":"a471defcdc02b94879c639539f1328915964e6f5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-06 12:23:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/15/check/gate-nova-pep8/f5ca32a/ : SUCCESS in 8m 48s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/15/check/gate-nova-docs/67e5e00//doc/build/html/ : SUCCESS in 14m 05s\n- gate-nova-python27 http://logs.openstack.org/35/193135/15/check/gate-nova-python27/0bf4c4b/ : FAILURE in 14m 33s\n- gate-nova-python34 http://logs.openstack.org/35/193135/15/check/gate-nova-python34/b2beb09/ : SUCCESS in 14m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-full/cae2ea4/ : FAILURE in 50m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-postgres-full/6703b38/ : FAILURE in 47m 57s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-neutron-full/067ba25/ : FAILURE in 1h 13m 45s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/15/check/gate-grenade-dsvm/41c0d98/ : SUCCESS in 50m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-large-ops/167c303/ : SUCCESS in 36m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-neutron-large-ops/729a948/ : SUCCESS in 29m 06s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/15/check/gate-nova-tox-functional/b4835d9/ : SUCCESS in 17m 31s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/15/check/gate-grenade-dsvm-multinode/10c1d8c/ : SUCCESS in 50m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c3f91e6/ : SUCCESS in 48m 25s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-nova-v20-api/0fb18b8/ : FAILURE in 42m 48s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-nova-v20-api-legacy/6095437/ : FAILURE in 16m 05s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-cells/1c205f3/ : FAILURE in 51m 25s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-full-ceph/4ea4a7a/ : FAILURE in 58m 06s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/15/check/gate-tempest-dsvm-multinode-full/65e20a2/ : FAILURE in 59m 50s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/15/check/gate-compute-api-guide/6580856//api-guide/build/html/ : SUCCESS in 8m 54s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/15/check/gate-nova-python27-constraints/c321322/ : FAILURE in 11m 14s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/15/check/gate-nova-python34-constraints/7b1493e/ : SUCCESS in 13m 18s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/15/check/gate-nova-docs-constraints/2fc3441//doc/build/html/ : SUCCESS in 13m 59s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/15/check/gate-nova-pep8-constraints/0c117af/ : SUCCESS in 9m 41s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/15/check/gate-nova-tox-constraints-functional-constraints/a249e55/ : SUCCESS in 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"df935ec1fa38f7dd66053caa0dca3fa1a87f3450","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-06 12:24:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/15/check/dsvm-tempest-xen/eb016e7 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":15},{"id":"19117dd756feccfffaabf72ffde2cb45f79d905b","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-07 15:05:10.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1262fc5667b05da14cafbd465bc7d39ea28a656d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-07 15:15:49.000000000","message":"Patch Set 16: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/16/18339/index.html : FAILURE in 05 m 09 s","accounts_in_message":[],"_revision_number":16},{"id":"09cabe4e49e8fda705300a1a9be3796fa35b096f","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-07 15:23:05.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"b0105418f47641dc5baca5d619acd89b58f8dfa3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-07 15:43:31.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/pci/193135/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"2bf5b062d26fcc03f9a49f418c6af430debb74cb","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-07 16:03:21.000000000","message":"Patch Set 17: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/17/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8b4a16233f6040a98a942347507e6520/index.html : SUCCESS in 15m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/17/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3391f12f534b40daaf6a451db5301d96/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/17/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/30295939feb14f7dbb584c0473f50289/index.html : SUCCESS in 13m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/17/check/real-db-upgrade_nova_percona_user_001:th-percona/84e4db3b5bf34eacb7ed4047a8fede00/index.html : SUCCESS in 16m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/17/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f7b1d758ecce4197a8d75e574fb3d213/index.html : SUCCESS in 15m 17s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/17/check/real-db-upgrade_nova_percona_user_002:th-percona/1f7b1996ec8849a39636db9586ccf006/index.html : SUCCESS in 12m 33s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":17},{"id":"7aa18094e7a543d8ccdb9be268214b84d32b2129","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-07 16:26:52.000000000","message":"Patch Set 17: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/17/18342/index.html : SUCCESS in 01 h 00 m 01 s","accounts_in_message":[],"_revision_number":17},{"id":"edbf977739e9893871a87fde08912ebe5fbb5cf0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-07 16:46:19.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/17/check/dsvm-tempest-xen/4319310 : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":17},{"id":"0e14f190dc49d79163c5994957cf71ff59c93b29","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-07 16:55:53.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/17/ : FAILURE in 1h 31m 44s","accounts_in_message":[],"_revision_number":17},{"id":"53eca189180f9f06dfbeacc71e39b3982077fe6f","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-07 17:45:00.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"83825f199f99d22de06b44e7b2d7aa50a9bb3046","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-07 18:22:04.000000000","message":"Patch Set 18: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/18/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/97ac86e4721c4f0ab8271c9d80861061/index.html : SUCCESS in 15m 07s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/18/check/real-db-upgrade_nova_mysql_user_001:th-mysql/817d9c4d2f57418a946e71bc319a2e73/index.html : SUCCESS in 13m 58s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/18/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3d5420d48920454c93cbc7d309ff15ea/index.html : SUCCESS in 15m 24s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/18/check/real-db-upgrade_nova_percona_user_001:th-percona/f61e72bdbea9483f889cbeac3cb99e34/index.html : SUCCESS in 17m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/18/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7168b5f08a8e4e5fa5484daae34d16c5/index.html : SUCCESS in 12m 57s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/18/check/real-db-upgrade_nova_percona_user_002:th-percona/2180c5d81f04499e9c99be12a4c31dad/index.html : SUCCESS in 15m 32s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":18},{"id":"e080575cc83b5fb8a63d4d8de9c21520c5edb244","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-07 18:38:59.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/18/check/check-ibm-tempest-dsvm-full/0cd4a36/ : FAILURE in 52m 42s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"9adf394d2ffab8b07a5f0209108a58fba9544629","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-07 18:50:02.000000000","message":"Patch Set 18: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/18/18376/index.html : FAILURE in 01 h 00 m 12 s","accounts_in_message":[],"_revision_number":18},{"id":"1a2cd6883b3ac11d4d0ca3f0d9bc76af2a206d0f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-07 18:54:35.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/18/ : FAILURE in 1h 00m 24s","accounts_in_message":[],"_revision_number":18},{"id":"cbadf9709cf139382b1dde4f4db92a7d338688f7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-07 18:58:28.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/193135/18 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"b8732cceca2ef3742c34313ad31a09168c1c1b45","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-07 19:14:56.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/18/check/dsvm-tempest-xen/714b9ec : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":18},{"id":"395b1756337ecdae17b6d6a4687f87ad92a21fef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-07 21:06:39.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/18/check/gate-nova-pep8/d21ee8d/ : SUCCESS in 8m 44s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/18/check/gate-nova-docs/55cf9b9//doc/build/html/ : SUCCESS in 14m 28s\n- gate-nova-python27 http://logs.openstack.org/35/193135/18/check/gate-nova-python27/9eafad4/ : FAILURE in 14m 09s\n- gate-nova-python34 http://logs.openstack.org/35/193135/18/check/gate-nova-python34/8134b01/ : SUCCESS in 13m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-full/8a82c99/ : FAILURE in 57m 28s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-postgres-full/1b3108d/ : FAILURE in 44m 06s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-neutron-full/ce71bdb/ : FAILURE in 1h 23m 33s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/18/check/gate-grenade-dsvm/e5e10a0/ : SUCCESS in 42m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-large-ops/0f04d0d/ : SUCCESS in 44m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-neutron-large-ops/b0cdc88/ : SUCCESS in 38m 24s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/18/check/gate-nova-tox-functional/0e98276/ : FAILURE in 16m 14s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/18/check/gate-grenade-dsvm-multinode/f5642db/ : FAILURE in 51m 59s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2e1947b/ : SUCCESS in 50m 14s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-nova-v20-api/17be26b/ : FAILURE in 42m 55s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-nova-v20-api-legacy/cc0f06f/ : FAILURE in 23m 21s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-cells/cf4cff7/ : FAILURE in 51m 59s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-full-ceph/330ea6c/ : FAILURE in 1h 01m 31s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/18/check/gate-tempest-dsvm-multinode-full/cbf003d/ : FAILURE in 1h 10m 02s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/18/check/gate-compute-api-guide/94fe8df//api-guide/build/html/ : SUCCESS in 8m 12s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/18/check/gate-nova-python27-constraints/47f89e2/ : FAILURE in 14m 30s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/18/check/gate-nova-python34-constraints/0d8ef39/ : SUCCESS in 13m 49s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/18/check/gate-nova-docs-constraints/e1f5871//doc/build/html/ : SUCCESS in 13m 03s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/18/check/gate-nova-pep8-constraints/b0a5c3d/ : SUCCESS in 9m 35s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/18/check/gate-nova-tox-constraints-functional-constraints/253b287/ : FAILURE in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a57fb0fa1c5915a5bbc38c48270f3d02e6647686","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-07 22:05:53.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"1f5ebc0b4491a20b1c77cecf9f441f5025661a6c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-07 22:39:50.000000000","message":"Patch Set 19: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/19/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2c105e9ad7424a538a09c299acd72f04/index.html : SUCCESS in 13m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/19/check/real-db-upgrade_nova_mysql_user_001:th-mysql/690476b7109e42c5b917eb4f2f0610a0/index.html : SUCCESS in 18m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/19/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7651d1d597674966b48f67e19999b34c/index.html : SUCCESS in 13m 03s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/19/check/real-db-upgrade_nova_percona_user_001:th-percona/919c976b732748c6af975fbb414a1f29/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/19/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4f7acbff53d3431390e72f0ed299e431/index.html : SUCCESS in 15m 58s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/19/check/real-db-upgrade_nova_percona_user_002:th-percona/7dcf68d86e664d9b9e7bcb0fc08cb8ce/index.html : SUCCESS in 16m 11s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":19},{"id":"b5e314b72a125bd1cf1b172edddf0424250a8f32","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-07 22:59:09.000000000","message":"Patch Set 19:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/19/check/check-ibm-tempest-dsvm-full/217a48b/ : FAILURE in 52m 11s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"994d4f7cee67cb614955c64adb6b0228f675e711","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-07 23:11:04.000000000","message":"Patch Set 19: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/19/18418/index.html : FAILURE in 01 h 00 m 03 s","accounts_in_message":[],"_revision_number":19},{"id":"8eafa2c42b6d3a44b27b2928bbd162fb93f23d46","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-08 00:01:01.000000000","message":"Patch Set 19:\n\n* pci-test http://52.27.155.124/pci/193135/19 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"aae149d5bf470d57d2a18e45870c3ea44088fd7b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-08 00:19:07.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/19/check/dsvm-tempest-xen/bcd38c4 : FAILURE in 1h 21m 34s","accounts_in_message":[],"_revision_number":19},{"id":"169efbe25a657273fbe06dca14a6b22994f76a92","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-08 00:29:27.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/19/ : FAILURE in 1h 40m 16s","accounts_in_message":[],"_revision_number":19},{"id":"81a12f9a24c2f6d45a40b3a6566eb1139c2831e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 01:19:06.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/19/check/gate-nova-pep8/10e9f58/ : SUCCESS in 8m 27s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/19/check/gate-nova-docs/efd1fd1//doc/build/html/ : SUCCESS in 12m 26s\n- gate-nova-python27 http://logs.openstack.org/35/193135/19/check/gate-nova-python27/7fe521e/ : FAILURE in 13m 00s\n- gate-nova-python34 http://logs.openstack.org/35/193135/19/check/gate-nova-python34/5322978/ : SUCCESS in 12m 32s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-full/80833df/ : FAILURE in 47m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-postgres-full/973f36b/ : FAILURE in 46m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-neutron-full/dcc5dcf/ : FAILURE in 1h 04m 06s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/19/check/gate-grenade-dsvm/1bb2df2/ : SUCCESS in 44m 56s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-large-ops/daee0a4/ : SUCCESS in 29m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-neutron-large-ops/8d011fd/ : SUCCESS in 28m 21s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/19/check/gate-nova-tox-functional/f9b25ca/ : FAILURE in 16m 13s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/19/check/gate-grenade-dsvm-multinode/b4df782/ : SUCCESS in 1h 15m 21s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2f57829/ : SUCCESS in 46m 28s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-nova-v20-api/93cf0d9/ : FAILURE in 34m 52s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-nova-v20-api-legacy/73580c0/ : FAILURE in 37m 51s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-cells/aede718/ : FAILURE in 52m 11s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-full-ceph/0c7512f/ : FAILURE in 47m 49s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/19/check/gate-tempest-dsvm-multinode-full/3df7dc5/ : FAILURE in 1h 18m 17s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/19/check/gate-compute-api-guide/b2f2c8e//api-guide/build/html/ : SUCCESS in 7m 04s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/19/check/gate-nova-python27-constraints/f316f71/ : FAILURE in 13m 49s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/19/check/gate-nova-python34-constraints/5fecad6/ : SUCCESS in 11m 54s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/19/check/gate-nova-docs-constraints/00b02bf//doc/build/html/ : SUCCESS in 12m 52s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/19/check/gate-nova-pep8-constraints/a4d9f07/ : SUCCESS in 8m 23s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/19/check/gate-nova-tox-constraints-functional-constraints/6a87c63/ : FAILURE in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"bb6626218fd2af7020469328812101da2add5849","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-08 13:36:40.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"93a03198cef0183f57980574972e69b2fc9a87e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-08 14:25:00.000000000","message":"Patch Set 20:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/20/check/check-ibm-tempest-dsvm-full/36ea953/ : FAILURE in 47m 06s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"22cd77ef1e0b1985e3ad96b55bfab3d5edc0b192","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-08 14:36:54.000000000","message":"Patch Set 20: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/20/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/01cc56b0e6b44364aafe2dc978a6e625/index.html : SUCCESS in 12m 45s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/20/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2e226379b13d4d22a658855296251d7b/index.html : SUCCESS in 14m 52s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/20/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5879d0fb2cf54d90beb0e353ec18d057/index.html : SUCCESS in 16m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/20/check/real-db-upgrade_nova_percona_user_001:th-percona/1d07359fc1f9424b9c00efabd34a7b59/index.html : SUCCESS in 17m 55s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/20/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4bfe36b8315249ce90badd26331fce99/index.html : SUCCESS in 19m 34s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/20/check/real-db-upgrade_nova_percona_user_002:th-percona/5f05a584465549e081ce1ab6824a7351/index.html : SUCCESS in 16m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":20},{"id":"940e55d3cc81b926abc9e5aed22eea5ff668b3c7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-08 14:53:28.000000000","message":"Patch Set 20: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/20/18556/index.html : FAILURE in 56 m 33 s","accounts_in_message":[],"_revision_number":20},{"id":"7c77962638fbc20b511b1d226c8782d4c56cfc97","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-08 15:25:14.000000000","message":"Patch Set 20:\n\n* pci-test http://52.27.155.124/pci/193135/20 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"ad99d9245b724cda8fb502086ee7ff459eaa0c83","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-08 15:25:39.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/20/check/dsvm-tempest-xen/53d871d : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":20},{"id":"3f34df1a589eb828ea8c586c5f91dd7c7d7ac27c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-08 16:10:57.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/20/ : FAILURE in 1h 43m 45s","accounts_in_message":[],"_revision_number":20},{"id":"f6795b3b73c94b0cc56a31e92fc40d1f78768b12","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-08 16:48:10.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6357ff90a9677223eb147b017436ffabd97cfe88","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-08 17:36:55.000000000","message":"Patch Set 21:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/21/check/check-ibm-tempest-dsvm-full/bffa3dd/ : FAILURE in 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"730ac3fa2fc2e785204302160f141f4bc58d3075","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-08 17:54:20.000000000","message":"Patch Set 21: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/21/18609/index.html : FAILURE in 55 m 33 s","accounts_in_message":[],"_revision_number":21},{"id":"d0a04563f0aa900c888da068ca0c2391432abfe3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-08 19:26:15.000000000","message":"Patch Set 21: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/21/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/719085473bae4048a21e7e4512579b1d/index.html : SUCCESS in 14m 26s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/21/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6af48f6dab25401b84b2183eb0eb88e0/index.html : SUCCESS in 16m 33s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/21/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dcf7ce884ea5468ba8d53e5985357126/index.html : SUCCESS in 14m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/21/check/real-db-upgrade_nova_percona_user_001:th-percona/62652b06e31d4ad6a617ad7034bc0f4a/index.html : SUCCESS in 16m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/21/check/real-db-upgrade_nova_mysql_user_002:th-mysql/cbef45903548443187f8798389afb7b4/index.html : SUCCESS in 12m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/21/check/real-db-upgrade_nova_percona_user_002:th-percona/84083c17a8fa4f89934bbe9ae025e37c/index.html : SUCCESS in 12m 57s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":21},{"id":"32c2851f83bdd6449b1b388c40f026b2a28b468e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 19:48:42.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/21/check/gate-nova-pep8/b688faf/ : SUCCESS in 10m 07s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/21/check/gate-nova-docs/26d1dec//doc/build/html/ : SUCCESS in 14m 33s\n- gate-nova-python27 http://logs.openstack.org/35/193135/21/check/gate-nova-python27/e04f30b/ : SUCCESS in 14m 29s\n- gate-nova-python34 http://logs.openstack.org/35/193135/21/check/gate-nova-python34/2b72a22/ : SUCCESS in 13m 21s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-full/22af69b/ : FAILURE in 1h 13m 12s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-postgres-full/91a4e4a/ : FAILURE in 47m 33s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-neutron-full/bc2aeea/ : FAILURE in 1h 17m 33s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/21/check/gate-grenade-dsvm/4aeeb33/ : SUCCESS in 52m 23s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-large-ops/fa9884d/ : SUCCESS in 30m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-neutron-large-ops/a94f7e8/ : SUCCESS in 36m 41s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/21/check/gate-nova-tox-functional/2b0e424/ : SUCCESS in 17m 17s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/21/check/gate-grenade-dsvm-multinode/7c53fce/ : SUCCESS in 1h 07m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8ca0819/ : SUCCESS in 53m 24s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-nova-v20-api/ecadbb2/ : FAILURE in 39m 16s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-nova-v20-api-legacy/de00dd7/ : FAILURE in 52m 31s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-cells/6469785/ : FAILURE in 1h 04m 47s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-full-ceph/f561881/ : FAILURE in 1h 05m 16s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/21/check/gate-tempest-dsvm-multinode-full/23e6f8b/ : FAILURE in 59m 44s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/21/check/gate-compute-api-guide/5bb9ff0//api-guide/build/html/ : SUCCESS in 9m 43s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/21/check/gate-nova-python27-constraints/0706952/ : SUCCESS in 14m 31s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/21/check/gate-nova-python34-constraints/2cd6f17/ : SUCCESS in 15m 37s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/21/check/gate-nova-docs-constraints/960aec5//doc/build/html/ : SUCCESS in 15m 13s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/21/check/gate-nova-pep8-constraints/8bd448e/ : SUCCESS in 10m 46s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/21/check/gate-nova-tox-constraints-functional-constraints/ed4c1b1/ : SUCCESS in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"103482831c64236231640d6d8e5ec6e477f90533","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-08 20:56:08.000000000","message":"Patch Set 21:\n\n* pci-test http://52.27.155.124/pci/193135/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"98de687294b904c80562b8a7d4ff8d13db5f8cec","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-08 21:15:40.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/21/check/dsvm-tempest-xen/d3a6036 : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":21},{"id":"28c1393dc93315efbc19531ec96a8e42bec8e54c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-08 21:20:33.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/21/ : FAILURE in 1h 35m 07s","accounts_in_message":[],"_revision_number":21},{"id":"48abf4dd5e4af340903b111b26a5e8c116c852b5","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-09 14:20:31.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"c440b01f3e92b9688d373a6bf5a6f0da255accda","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-09 15:13:01.000000000","message":"Patch Set 22:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/22/check/check-ibm-tempest-dsvm-full/0e2eece/ : FAILURE in 51m 13s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"daf474eb2c07aaae30dc1212e64303ad3d3dc0fc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-09 15:15:50.000000000","message":"Patch Set 22: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/22/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5f4619f849f540f9b5f1895301666d39/index.html : SUCCESS in 14m 33s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/22/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5bc5aa17ebe94bb7b83297c1d1cbbad6/index.html : SUCCESS in 15m 41s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/22/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1be31053e5c84900b3c88808e3580c2c/index.html : SUCCESS in 13m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/22/check/real-db-upgrade_nova_percona_user_001:th-percona/9101770293604aeb891f6a66af9110c1/index.html : SUCCESS in 18m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/22/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f075c51d5b454636bdbd0d15ab321fe5/index.html : SUCCESS in 14m 44s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/22/check/real-db-upgrade_nova_percona_user_002:th-percona/2e86bb55e0764e12b97d19eaadcc054c/index.html : SUCCESS in 12m 39s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":22},{"id":"bd26c0b43001116ba98b0a1eefee1a9f81c7066e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-09 15:34:15.000000000","message":"Patch Set 22: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/22/18792/index.html : FAILURE in 01 h 09 m 58 s","accounts_in_message":[],"_revision_number":22},{"id":"e526467e55d57b9bd0b20669ab49df2ccf87de4f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-09 15:35:30.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/22/check/dsvm-tempest-xen/8b1b4b3 : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":22},{"id":"4ce0adf03d3ac8a04521f36af5e43bc7b90ce075","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-09 15:55:41.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/193135/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"c3c82a8d89c537d1e8a7eca64ee65a25bf4671e7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-09 16:01:16.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/22/Nova-ML2-Sriov : SUCCESS in 55m 33s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"a0164293c427d3bd6a85f83e5c568e9a5c83cbc6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-09 16:02:58.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/22/ : FAILURE in 1h 24m 51s","accounts_in_message":[],"_revision_number":22},{"id":"207fe6881181e4764c4e5c7d23d4c4d19eb425bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-09 16:06:41.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/22/check/gate-nova-pep8/8f678db/ : SUCCESS in 7m 11s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/22/check/gate-nova-docs/79c9aa2//doc/build/html/ : SUCCESS in 15m 09s\n- gate-nova-python27 http://logs.openstack.org/35/193135/22/check/gate-nova-python27/bc5c379/ : SUCCESS in 16m 20s\n- gate-nova-python34 http://logs.openstack.org/35/193135/22/check/gate-nova-python34/8ea4af1/ : SUCCESS in 12m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-full/e7bea47/ : FAILURE in 1h 11m 28s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-postgres-full/8e5d1fd/ : FAILURE in 55m 14s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-neutron-full/495ae5a/ : FAILURE in 1h 30m 34s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/22/check/gate-grenade-dsvm/2ece158/ : SUCCESS in 47m 15s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-large-ops/87ac513/ : SUCCESS in 42m 40s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-neutron-large-ops/5ec222c/ : SUCCESS in 36m 08s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/22/check/gate-nova-tox-functional/8daab9b/ : SUCCESS in 18m 27s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/22/check/gate-grenade-dsvm-multinode/da6670c/ : SUCCESS in 48m 37s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/da96c4a/ : SUCCESS in 50m 28s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-nova-v20-api/6c75cd6/ : FAILURE in 48m 53s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-nova-v20-api-legacy/fdcecab/ : FAILURE in 42m 49s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-cells/53e035f/ : FAILURE in 50m 22s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-full-ceph/bd86716/ : FAILURE in 1h 06m 13s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/22/check/gate-tempest-dsvm-multinode-full/34af3e0/ : FAILURE in 1h 09m 46s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/22/check/gate-compute-api-guide/fcd4462//api-guide/build/html/ : SUCCESS in 6m 59s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/22/check/gate-nova-python27-constraints/15a485d/ : SUCCESS in 14m 15s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/22/check/gate-nova-python34-constraints/53066df/ : SUCCESS in 11m 25s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/22/check/gate-nova-docs-constraints/a68c83c//doc/build/html/ : SUCCESS in 15m 31s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/22/check/gate-nova-pep8-constraints/202ecd1/ : SUCCESS in 8m 21s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/22/check/gate-nova-tox-constraints-functional-constraints/8ba52c9/ : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"bd68f6f04500214f74e6d18becb58e3c9550489d","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-09 19:27:21.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"06d27b2364793389ded54a495c37c7a42a1fe5aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-09 19:28:07.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"1b0fd429c9d9dfce60ed3309ee932cca8b4c0cc4","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-09 19:28:17.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"4c21c8651c74626aa39d971e882769661e431379","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-09 19:28:24.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"034c99cf54818c3cc45e0023eeac1b0873ef5469","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-09 19:28:30.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":23},{"id":"3a9bd230b33de11d758a2c4c46584b76b6016863","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-09 19:29:00.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":23},{"id":"f8d378729e533cae54c8a806fa6572cddd8573f0","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-09 19:35:17.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"dba148954a3b486bbb3678b60ab94787361623eb","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-09 20:18:16.000000000","message":"Patch Set 24: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/24/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/357b5323d239469f8ab68b18fc6dd4e4/index.html : SUCCESS in 14m 45s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/24/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f70dea0d56b04b3183b9a34dd643d8b9/index.html : SUCCESS in 16m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/24/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/eef360c03b744d92888551936ce9fb7d/index.html : SUCCESS in 13m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/24/check/real-db-upgrade_nova_percona_user_001:th-percona/5006fb5028cf4bfb95526ad469beec68/index.html : SUCCESS in 17m 21s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/24/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6e2256f2d78c407a9cf646e167f8eac3/index.html : SUCCESS in 12m 51s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/24/check/real-db-upgrade_nova_percona_user_002:th-percona/3709d3dc687344c3ae677f558d41bdb5/index.html : SUCCESS in 15m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":24},{"id":"6abe6bd3ba1a32e10ce4d94ba2e50f3e6d89cc6d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-09 20:31:57.000000000","message":"Patch Set 24:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/24/check/check-ibm-tempest-dsvm-full/5bca458/ : FAILURE in 55m 21s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"1cb4302a0ce62a7fee2e2eb0ee66a23e0fb4f5f2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-09 20:35:40.000000000","message":"Patch Set 24:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/24/ : FAILURE in 58m 33s","accounts_in_message":[],"_revision_number":24},{"id":"b3ff417bdf1c29e22a0f813cf34ac3296aaaf88c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-09 20:36:16.000000000","message":"Patch Set 24: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/24/18847/index.html : FAILURE in 56 m 40 s","accounts_in_message":[],"_revision_number":24},{"id":"a8c68fdfc763452a8851ca3760fc7e9a6b54f9f1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-09 20:52:01.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/24/check/dsvm-tempest-xen/4a0c2eb : SUCCESS in 1h 15m 30s","accounts_in_message":[],"_revision_number":24},{"id":"a936d3f0d1db6d2567858df551c36059ed860e20","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-09 21:29:53.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/193135/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"27d78a26c0b68fe876190080bca67a03f891766b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-09 21:47:20.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/pci/193135/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"2d4ae43ecfaef5653084db36f3cd3b1b3c801839","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-09 21:53:42.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/24/check/gate-nova-pep8/eb919d3/ : SUCCESS in 8m 50s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/24/check/gate-nova-docs/92c57ed//doc/build/html/ : SUCCESS in 23m 55s\n- gate-nova-python27 http://logs.openstack.org/35/193135/24/check/gate-nova-python27/e8b22d7/ : SUCCESS in 16m 04s\n- gate-nova-python34 http://logs.openstack.org/35/193135/24/check/gate-nova-python34/5e43b1d/ : SUCCESS in 14m 22s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-full/deb5ca3/ : FAILURE in 1h 08m 42s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-postgres-full/abce2b1/ : FAILURE in 1h 04m 12s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-neutron-full/3595084/ : FAILURE in 1h 36m 26s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/24/check/gate-grenade-dsvm/687b081/ : FAILURE in 54m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-large-ops/fce42ea/ : SUCCESS in 34m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-neutron-large-ops/5e34dd5/ : SUCCESS in 34m 13s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/24/check/gate-nova-tox-functional/5d5c9ba/ : SUCCESS in 19m 11s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/24/check/gate-grenade-dsvm-multinode/f5e9221/ : FAILURE in 51m 34s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/dca9346/ : SUCCESS in 40m 52s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-nova-v20-api/17e9092/ : FAILURE in 33m 00s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-nova-v20-api-legacy/d5fa09d/ : FAILURE in 40m 46s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-cells/9293350/ : FAILURE in 50m 09s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-full-ceph/ba33394/ : FAILURE in 55m 55s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/24/check/gate-tempest-dsvm-multinode-full/499a76a/ : FAILURE in 1h 03m 38s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/24/check/gate-compute-api-guide/6882bcb//api-guide/build/html/ : SUCCESS in 8m 52s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/24/check/gate-nova-python27-constraints/4550b4c/ : SUCCESS in 15m 45s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/24/check/gate-nova-python34-constraints/9d8e277/ : SUCCESS in 13m 14s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/24/check/gate-nova-docs-constraints/f0e85fd//doc/build/html/ : SUCCESS in 14m 10s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/24/check/gate-nova-pep8-constraints/80a3f97/ : SUCCESS in 10m 11s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/24/check/gate-nova-tox-constraints-functional-constraints/2d77881/ : SUCCESS in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"ea7c89837c0fd18f1969551a9f4310b22230a6b0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-09 22:02:07.000000000","message":"Patch Set 24:\n\n* pci-test http://52.27.155.124/pci/193135/24 : SUCCESS","accounts_in_message":[],"_revision_number":24},{"id":"f007790cc6b85b7f7a0dfd357331b42e94bb6caa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-10 00:34:40.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/24/Nova-ML2-Sriov : SUCCESS in 1h 00m 48s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":24},{"id":"ab2d46fc9959ecdf39d008b0dd42de771d68205b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-10 15:56:48.000000000","message":"Patch Set 20:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/20 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/20 : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"2bc050873ccbe97173858d8621a1a5fd0ff6fc53","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-11 07:08:35.000000000","message":"Patch Set 24:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/24 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"442cb997ff267b6bf9548b477453fd3703b363ce","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-11 09:21:12.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"ccca2058239af55fea6d8109ffc0a169d9512f1f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-11 09:21:54.000000000","message":"Patch Set 25:\n\n* pci-test http://52.27.155.124/pci/193135/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"45e8005df0a95ad8b79010e9ec4ba0658b1f6156","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-11 09:28:25.000000000","message":"Patch Set 25:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/25/check/check-ibm-tempest-dsvm-full/d69fa8f/ : FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b8a516318014a36c4139fd754e66c25cfc6e9a66","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-11 09:30:43.000000000","message":"Patch Set 25: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/25/19167/index.html : FAILURE in 07 m 17 s","accounts_in_message":[],"_revision_number":25},{"id":"4faaaa6632dc0c8ae1ab52f8cf48cdabd5f88589","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-11 09:31:43.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/25/check/dsvm-tempest-xen/9183653 : FAILURE in 9m 29s","accounts_in_message":[],"_revision_number":25},{"id":"9161889c021453f508386f6708af5b3ac6f3e356","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-11 10:09:53.000000000","message":"Patch Set 25: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/25/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/ef9d6e37f2794f64a5350bfab4901ef3/index.html : SUCCESS in 12m 48s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/25/check/real-db-upgrade_nova_mysql_user_001:th-mysql/763672a6209547d4b761fb3082b58828/index.html : SUCCESS in 15m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/25/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a97a0d12b0ac416e884be61f0a3cf9d2/index.html : SUCCESS in 13m 01s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/25/check/real-db-upgrade_nova_percona_user_001:th-percona/439b66beb3b84e0bb329ac49b1f2aa28/index.html : SUCCESS in 13m 53s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/25/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e43f920366fc436b814edebb37ed2405/index.html : SUCCESS in 15m 46s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/25/check/real-db-upgrade_nova_percona_user_002:th-percona/8734fb4a99554436920f94d652eeb56d/index.html : SUCCESS in 14m 27s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":25},{"id":"d0887837aea0de14cfc8942bba19d9b3f1dba60d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-11 10:44:50.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/25/check/gate-nova-pep8/d9afbc2/ : SUCCESS in 9m 39s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/25/check/gate-nova-docs/ded88ee//doc/build/html/ : SUCCESS in 14m 20s\n- gate-nova-python27 http://logs.openstack.org/35/193135/25/check/gate-nova-python27/e4633b1/ : SUCCESS in 17m 09s\n- gate-nova-python34 http://logs.openstack.org/35/193135/25/check/gate-nova-python34/fba45a8/ : SUCCESS in 12m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-full/a673f5a/ : FAILURE in 9m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-postgres-full/939350d/ : FAILURE in 10m 08s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-neutron-full/76d2c6e/ : FAILURE in 9m 52s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/25/check/gate-grenade-dsvm/ac02cec/ : FAILURE in 43m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-large-ops/e25a73b/ : FAILURE in 9m 35s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-neutron-large-ops/06c781c/ : FAILURE in 9m 39s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/25/check/gate-nova-tox-functional/98f864c/ : SUCCESS in 16m 54s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/25/check/gate-grenade-dsvm-multinode/41fa25c/ : FAILURE in 42m 11s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/286a850/ : FAILURE in 9m 41s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-nova-v20-api/699bcd3/ : FAILURE in 12m 33s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-nova-v20-api-legacy/afe2932/ : FAILURE in 6m 30s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-cells/4e83e8a/ : FAILURE in 10m 45s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-full-ceph/70101cd/ : FAILURE in 12m 12s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/25/check/gate-tempest-dsvm-multinode-full/c6151f5/ : FAILURE in 13m 13s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/25/check/gate-compute-api-guide/5cb3c82//api-guide/build/html/ : SUCCESS in 6m 45s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/25/check/gate-nova-python27-constraints/63f59b1/ : FAILURE in 2m 31s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/25/check/gate-nova-python34-constraints/3a04611/ : FAILURE in 1m 45s (non-voting)\n- gate-nova-docs-constraints http://logs.openstack.org/35/193135/25/check/gate-nova-docs-constraints/6621bcd/ : FAILURE in 2m 32s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/25/check/gate-nova-pep8-constraints/279cfa7/ : FAILURE in 3m 03s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/25/check/gate-nova-tox-constraints-functional-constraints/c8a7855/ : FAILURE in 2m 32s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"0486c0f63d008507b471b2dc0da35e43db88fb3a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-11 10:49:45.000000000","message":"Patch Set 25:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/25/ : FAILURE in 1h 26m 56s","accounts_in_message":[],"_revision_number":25},{"id":"b1f76b10c013e06b2bc8d9917d849164f3ce861b","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-11 20:45:38.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"43c38dff1046ce2ba154f1d0fab9dfbbd741568e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-11 21:05:45.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/26/check/dsvm-tempest-xen/6bd7295 : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":26},{"id":"6d455b0fedc9e303f8df4ff6ea9935a2da8cac39","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-11 21:41:04.000000000","message":"Patch Set 26:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/26/check/check-ibm-tempest-dsvm-full/af9b6eb/ : FAILURE in 50m 31s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"686ba2f690e855cc25d12d59cb9b087d6073699c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-11 21:41:43.000000000","message":"Patch Set 26: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/26/19286/index.html : SUCCESS in 52 m 43 s","accounts_in_message":[],"_revision_number":26},{"id":"403a32d6a68a54ab9a3cc0b08784dad63e31d1df","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-11 22:27:08.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/26/ : SUCCESS in 1h 28m 46s","accounts_in_message":[],"_revision_number":26},{"id":"85f825e482e94b398f4d853caa771eee4ecbea10","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-11 22:40:11.000000000","message":"Patch Set 26: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/26/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9246fd7c355a455286ff003e1328c83d/index.html : SUCCESS in 15m 01s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/26/check/real-db-upgrade_nova_mysql_user_001:th-mysql/af252b8aa7064cd8b361d0e2b794fdaf/index.html : SUCCESS in 14m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/26/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2f145ae4bd094638a50b0af95f3c3146/index.html : SUCCESS in 14m 32s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/26/check/real-db-upgrade_nova_percona_user_001:th-percona/9fcde150e6344a2c83ed31a7a4ff5959/index.html : SUCCESS in 17m 07s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/26/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1c0d1c757f8646b5a3db43cb6c0e0ee6/index.html : SUCCESS in 13m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/26/check/real-db-upgrade_nova_percona_user_002:th-percona/42eb93a10fde4e99b992ed08365bd67a/index.html : SUCCESS in 15m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":26},{"id":"ef2585f1b34341a134c7912b76dbba9dd26cd3c6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-11 22:45:46.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/26/check/gate-nova-pep8/111808d/ : SUCCESS in 6m 52s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/26/check/gate-nova-docs/d06bf5b//doc/build/html/ : SUCCESS in 13m 52s\n- gate-nova-python27 http://logs.openstack.org/35/193135/26/check/gate-nova-python27/84eaadd/ : SUCCESS in 13m 47s\n- gate-nova-python34 http://logs.openstack.org/35/193135/26/check/gate-nova-python34/9facac5/ : SUCCESS in 12m 48s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-full/32353da/ : FAILURE in 58m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-postgres-full/85baa41/ : FAILURE in 55m 40s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-neutron-full/90bfb58/ : FAILURE in 1h 15m 15s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/26/check/gate-grenade-dsvm/49d682e/ : SUCCESS in 51m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-large-ops/b49e2f6/ : SUCCESS in 40m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-neutron-large-ops/faad795/ : SUCCESS in 34m 52s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/26/check/gate-nova-tox-functional/71dc753/ : SUCCESS in 17m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/26/check/gate-grenade-dsvm-multinode/7e6494b/ : SUCCESS in 59m 12s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/fb4cb8d/ : SUCCESS in 45m 41s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-nova-v20-api/973c6a3/ : SUCCESS in 50m 20s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-nova-v20-api-legacy/2e4195f/ : SUCCESS in 33m 13s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-cells/7404562/ : SUCCESS in 46m 18s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-full-ceph/c8c7333/ : FAILURE in 1h 01m 12s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/26/check/gate-tempest-dsvm-multinode-full/450ac8f/ : FAILURE in 1h 00m 19s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/26/check/gate-compute-api-guide/d5981f6//api-guide/build/html/ : SUCCESS in 8m 41s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/26/check/gate-nova-python27-constraints/e83f30e/ : SUCCESS in 14m 38s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/26/check/gate-nova-python34-constraints/9c07688/ : SUCCESS in 12m 48s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/26/check/gate-nova-docs-constraints/9d7a32e//doc/build/html/ : SUCCESS in 9m 12s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/26/check/gate-nova-pep8-constraints/375d172/ : SUCCESS in 9m 10s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/26/check/gate-nova-tox-constraints-functional-constraints/594ad61/ : SUCCESS in 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"7c8bd312195c259d8d5a208a0a369fbcdd837ff3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-12 00:31:52.000000000","message":"Patch Set 26:\n\n* pci-test http://52.27.155.124/pci/193135/26 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"1673842e0fa2d63ea3f5629ffbf7bcf0fa548e1f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-13 02:12:26.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/26/Nova-ML2-Sriov : FAILURE in 56m 20s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":26},{"id":"303962fa214e22548ff8dda2dbb9b4e790c081f2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-13 03:33:32.000000000","message":"Patch Set 26: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/26 \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/26 : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"33628bb0a7f1aa0f32cd5344f3d1239ef399c057","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-13 11:11:05.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"7ff8b5ca26e1b44a6c0a46e0b289c81674baed33","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-13 11:28:29.000000000","message":"Patch Set 27: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/27/19329/index.html : FAILURE in 15 m 34 s","accounts_in_message":[],"_revision_number":27},{"id":"16131160a638c964109ca76c8f108440f9dad1f5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-13 11:28:40.000000000","message":"Patch Set 27:\n\n* pci-test http://52.27.155.124/pci/193135/27 : SUCCESS","accounts_in_message":[],"_revision_number":27},{"id":"273d80b36f1283c23dbd44b5e760270f32e86a0f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-13 11:30:12.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/27/check/dsvm-tempest-xen/74a08f1 : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":27},{"id":"edd2917ea27c66650792a4696d8e41dce22b2289","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-13 12:01:47.000000000","message":"Patch Set 27: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/27 \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/27 : SUCCESS","accounts_in_message":[],"_revision_number":27},{"id":"288b07285a6c2f14fd2faa2d35a2fbf27dfe00aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-13 12:24:08.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/27/check/gate-nova-pep8/4eda8d0/ : FAILURE in 7m 01s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/27/check/gate-nova-docs/b7163f9//doc/build/html/ : SUCCESS in 16m 23s\n- gate-nova-python27 http://logs.openstack.org/35/193135/27/check/gate-nova-python27/89ebb7d/ : FAILURE in 11m 16s\n- gate-nova-python34 http://logs.openstack.org/35/193135/27/check/gate-nova-python34/ea85bf6/ : SUCCESS in 10m 24s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-full/4e46bd1/ : FAILURE in 59m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-postgres-full/34c80af/ : FAILURE in 55m 29s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-neutron-full/8a8496b/ : FAILURE in 59m 49s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/27/check/gate-grenade-dsvm/dd0ecc7/ : SUCCESS in 44m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-large-ops/156e978/ : SUCCESS in 32m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-neutron-large-ops/28c11fa/ : SUCCESS in 36m 07s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/27/check/gate-nova-tox-functional/1b52442/ : SUCCESS in 16m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/27/check/gate-grenade-dsvm-multinode/964a44d/ : SUCCESS in 49m 26s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/42e15b1/ : SUCCESS in 42m 19s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-nova-v20-api/d1e57e2/ : SUCCESS in 42m 47s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-nova-v20-api-legacy/22a3f8d/ : SUCCESS in 38m 38s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-cells/4ec68a2/ : SUCCESS in 46m 22s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-full-ceph/b26915b/ : FAILURE in 58m 27s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/a31bac5/ : FAILURE in 59m 30s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/27/check/gate-tempest-dsvm-multinode-full/4cc5622/ : FAILURE in 1h 06m 13s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/27/check/gate-compute-api-guide/3b47b36//api-guide/build/html/ : SUCCESS in 9m 15s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/27/check/gate-nova-python27-constraints/2e01995/ : FAILURE in 11m 27s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/27/check/gate-nova-python34-constraints/347ccd4/ : SUCCESS in 13m 20s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/27/check/gate-nova-docs-constraints/e083cd9//doc/build/html/ : SUCCESS in 14m 09s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/27/check/gate-nova-pep8-constraints/b940a2e/ : FAILURE in 9m 34s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/27/check/gate-nova-tox-constraints-functional-constraints/1d18417/ : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"44819c6d8053269aa778fc0849252626f18cda2d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-13 12:44:11.000000000","message":"Patch Set 27:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/27/ : FAILURE in 1h 31m 29s","accounts_in_message":[],"_revision_number":27},{"id":"ac0a8ce3f94fa150f082be33649cfeeb8d6c0ea0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-13 13:48:47.000000000","message":"Patch Set 27:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/27/check/check-ibm-tempest-dsvm-full/c40eb14/ : FAILURE in 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a800055cedd7540c64d95ee79e63f84d6762a19c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-13 14:27:36.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"472e1239e5648e80f83390af1998eb763c906c1d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-13 14:43:26.000000000","message":"Patch Set 28:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/28/check/check-ibm-tempest-dsvm-full/f8e90cf/ : FAILURE in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"506a906f85fbdd0c9712acb7337dfed682c89489","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-13 14:44:49.000000000","message":"Patch Set 28:\n\n* pci-test http://52.27.155.124/pci/193135/28 : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"ebbee36c76970888dfe3b1e5b6f40f058ea76c78","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-13 14:45:14.000000000","message":"Patch Set 28: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/28/19335/index.html : FAILURE in 16 m 11 s","accounts_in_message":[],"_revision_number":28},{"id":"6c435802746d172ee19c18d83d4d59652f01049a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-13 14:46:34.000000000","message":"Patch Set 28: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/28/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4de027c4cf404956939f5bbd82cdcae8/index.html : SUCCESS in 12m 28s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/28/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0e9b0a42520c433e98bb75f99149bc35/index.html : SUCCESS in 16m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/28/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4b2e96f6541f419191808cf151f80b05/index.html : SUCCESS in 15m 56s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/28/check/real-db-upgrade_nova_percona_user_001:th-percona/1e14e62a02be435fb41ca2d01be436b3/index.html : SUCCESS in 13m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/28/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d49cf90853d8450f84f1f7ca9ffdc10e/index.html : SUCCESS in 12m 58s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/28/check/real-db-upgrade_nova_percona_user_002:th-percona/3d8dae61c65f4bb3baad5f7b5c0fca85/index.html : SUCCESS in 13m 04s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":28},{"id":"f7960c5192a330f3a56d303c597e2e8236c1bec6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-13 14:47:16.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/28/check/dsvm-tempest-xen/4f9db2f : FAILURE in 18m 42s","accounts_in_message":[],"_revision_number":28},{"id":"c67fa33870618ed8c950cd7801b6641a7f27d5ae","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-13 15:18:50.000000000","message":"Patch Set 28: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/28 \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/28 : SUCCESS","accounts_in_message":[],"_revision_number":28},{"id":"f4196d874a4151a9c63624ead848035f0112201a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-13 15:33:53.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/28/check/gate-nova-pep8/b152766/ : SUCCESS in 7m 13s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/28/check/gate-nova-docs/e41c58e//doc/build/html/ : SUCCESS in 15m 04s\n- gate-nova-python27 http://logs.openstack.org/35/193135/28/check/gate-nova-python27/f90d248/ : SUCCESS in 12m 26s\n- gate-nova-python34 http://logs.openstack.org/35/193135/28/check/gate-nova-python34/53d6d6c/ : SUCCESS in 10m 27s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-full/01df46a/ : FAILURE in 1h 04m 22s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-postgres-full/d5880f7/ : FAILURE in 49m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-neutron-full/be8e14a/ : FAILURE in 1h 05m 08s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/28/check/gate-grenade-dsvm/f11c2d6/ : SUCCESS in 47m 54s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-large-ops/57189b3/ : SUCCESS in 39m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-neutron-large-ops/d35b0e5/ : SUCCESS in 39m 18s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/28/check/gate-nova-tox-functional/c46b1a9/ : SUCCESS in 14m 11s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/28/check/gate-grenade-dsvm-multinode/61ea552/ : SUCCESS in 49m 48s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f6736ff/ : SUCCESS in 44m 23s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-nova-v20-api/a4ac03a/ : SUCCESS in 41m 05s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-nova-v20-api-legacy/74237c9/ : SUCCESS in 36m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-cells/abe4334/ : SUCCESS in 36m 41s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-full-ceph/debd1b8/ : FAILURE in 52m 39s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5b29ed2/ : FAILURE in 50m 43s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/28/check/gate-tempest-dsvm-multinode-full/91238b6/ : FAILURE in 1h 01m 02s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/28/check/gate-compute-api-guide/0fc1dbc//api-guide/build/html/ : SUCCESS in 9m 20s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/28/check/gate-nova-python27-constraints/4031cfc/ : SUCCESS in 14m 42s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/28/check/gate-nova-python34-constraints/141d509/ : SUCCESS in 12m 46s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/28/check/gate-nova-docs-constraints/0000de7//doc/build/html/ : SUCCESS in 14m 56s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/28/check/gate-nova-pep8-constraints/61287cb/ : SUCCESS in 9m 06s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/28/check/gate-nova-tox-constraints-functional-constraints/5bdffbe/ : SUCCESS in 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"335cf9e6c463d433d4128a5b79c7c3c1e1b63ce5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-13 16:13:32.000000000","message":"Patch Set 28:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/28/ : FAILURE in 1h 44m 27s","accounts_in_message":[],"_revision_number":28},{"id":"1cc45b52241db7631b26fcdd04a08c81d8885641","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-13 16:17:44.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"a2528fafe2cc3c0a101dcc0f11eb94aad61885c0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-13 16:33:48.000000000","message":"Patch Set 29:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/29/check/check-ibm-tempest-dsvm-full/901d8a6/ : FAILURE in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"56e493543bc5e2be9119e25df325a30f1b451f66","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-13 16:35:12.000000000","message":"Patch Set 29:\n\n* pci-test http://52.27.155.124/pci/193135/29 : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"e0e02281a0d55932bd29d11326f61f069dba2499","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-13 16:36:27.000000000","message":"Patch Set 29: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/29/19340/index.html : FAILURE in 16 m 36 s","accounts_in_message":[],"_revision_number":29},{"id":"8104c2b04843b213fa97692890129a76d0edb5d9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-13 16:36:46.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/29/check/dsvm-tempest-xen/36e385e : FAILURE in 17m 51s","accounts_in_message":[],"_revision_number":29},{"id":"ff149bc64d31678a960a358db00ad79b5fb56ab6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-13 16:58:42.000000000","message":"Patch Set 29:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/29 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"d1cbabd2940d84b106012422c0305316e77533e9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-13 17:05:52.000000000","message":"Patch Set 29: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/29/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bf4f9de8007244928ddad34c451150db/index.html : SUCCESS in 14m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/29/check/real-db-upgrade_nova_mysql_user_001:th-mysql/359a949b766d41e49ccbfdca799583d7/index.html : SUCCESS in 14m 42s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/29/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1ae9bc2551bf43ae84bd929afb10a19e/index.html : SUCCESS in 14m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/29/check/real-db-upgrade_nova_percona_user_001:th-percona/3f9458bdc3dd432099d2a35c20df3863/index.html : SUCCESS in 16m 04s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/29/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4543f8f9a7ba46b0aa3fa73e8ce9803c/index.html : SUCCESS in 12m 28s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/29/check/real-db-upgrade_nova_percona_user_002:th-percona/a287b97561bf42f5979e46eb99a54e76/index.html : SUCCESS in 14m 36s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":29},{"id":"86be6674c0e132f259ca47a0901326c94537b0cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-13 17:29:59.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/29/check/gate-nova-pep8/b060c45/ : SUCCESS in 9m 21s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/29/check/gate-nova-docs/62894b3//doc/build/html/ : SUCCESS in 12m 24s\n- gate-nova-python27 http://logs.openstack.org/35/193135/29/check/gate-nova-python27/2c32526/ : SUCCESS in 15m 02s\n- gate-nova-python34 http://logs.openstack.org/35/193135/29/check/gate-nova-python34/3d1e7dd/ : SUCCESS in 9m 43s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-full/ebcacda/ : FAILURE in 49m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-postgres-full/bce9542/ : FAILURE in 49m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-neutron-full/5ad00d5/ : FAILURE in 1h 10m 43s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/29/check/gate-grenade-dsvm/da7641c/ : SUCCESS in 53m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-large-ops/8479d9b/ : SUCCESS in 31m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-neutron-large-ops/696e7e8/ : SUCCESS in 40m 19s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/29/check/gate-nova-tox-functional/41881d3/ : SUCCESS in 13m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/29/check/gate-grenade-dsvm-multinode/20e8e5e/ : SUCCESS in 55m 12s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/751228e/ : SUCCESS in 43m 14s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-nova-v20-api/2c4fece/ : SUCCESS in 42m 03s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-nova-v20-api-legacy/474929b/ : SUCCESS in 36m 50s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-cells/74c0a59/ : SUCCESS in 46m 14s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-full-ceph/32fdc3d/ : FAILURE in 50m 58s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4027d26/ : FAILURE in 1h 02m 13s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/29/check/gate-tempest-dsvm-multinode-full/2899352/ : FAILURE in 1h 01m 22s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/29/check/gate-compute-api-guide/3fd48f1//api-guide/build/html/ : SUCCESS in 6m 46s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/29/check/gate-nova-python27-constraints/dc96a66/ : SUCCESS in 11m 32s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/29/check/gate-nova-python34-constraints/4ac922a/ : SUCCESS in 12m 40s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/29/check/gate-nova-docs-constraints/976998b//doc/build/html/ : SUCCESS in 13m 59s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/29/check/gate-nova-pep8-constraints/db271bc/ : SUCCESS in 9m 09s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/29/check/gate-nova-tox-constraints-functional-constraints/4f8d9a6/ : SUCCESS in 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"a07a330656740a361e95c5098c7afc8067ee82cc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-13 17:38:22.000000000","message":"Patch Set 29:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/29/ : FAILURE in 1h 18m 38s","accounts_in_message":[],"_revision_number":29},{"id":"9367eff80ccb9e30e3e7251a10fc61349c6727ef","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-13 17:49:40.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/29/Nova-ML2-Sriov : SUCCESS in 56m 15s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":29},{"id":"de45df00497cf246475266714c2a992fa20514ed","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-15 06:03:06.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"73e91bdb338f8c0b1b30db668934075a76b47f7c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-15 06:59:06.000000000","message":"Patch Set 30:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/30/check/check-ibm-tempest-dsvm-full/fbfa752/ : FAILURE in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"d00bb3d3e1907eb20c0d25aefe3adebf8d3a26c0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-15 07:00:10.000000000","message":"Patch Set 30: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/30/19614/index.html : SUCCESS in 54 m 17 s","accounts_in_message":[],"_revision_number":30},{"id":"9fadf2ebfc58de00ff1487c640bef1dae597e18b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-15 07:10:57.000000000","message":"Patch Set 30: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/30/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6a0fa678614d492aaec6a1ae2ffce49b/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/30/check/real-db-upgrade_nova_mysql_user_001:th-mysql/82e2c1ac90a14200bbc329a1856cecd0/index.html : SUCCESS in 17m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/30/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d0ff1d8cdf3c414f9b5cc7f9e6ec3794/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/30/check/real-db-upgrade_nova_percona_user_001:th-percona/d3f8b7cfa79d440da903df47be12fa9c/index.html : SUCCESS in 13m 08s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://localhost : FAILURE during cleanup and log upload\nException: Authorization Failure. Authorization Failed: Unable to establish connection to https://identity.api.rackspacecloud.com/v2.0/tokens in 12m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/30/check/real-db-upgrade_nova_percona_user_002:th-percona/8112f924d79a4823a67b862bc0bb4db0/index.html : SUCCESS in 15m 59s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":30},{"id":"7bd0fba4c6df52b183c8ebcc08d70d3c6a86ed03","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-15 07:16:50.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/30/check/dsvm-tempest-xen/c82b44a : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":30},{"id":"56fb0e86f3926d9994b2d5daf23b2fc04d2d9b68","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-15 07:29:20.000000000","message":"Patch Set 30:\n\n* pci-test http://52.27.155.124/pci/193135/30 : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"3d8efdba948757880022a37e84007c31bfac42ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 07:34:50.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/30/check/gate-nova-pep8/ee5e810/ : SUCCESS in 8m 47s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/30/check/gate-nova-docs/f2b0508//doc/build/html/ : SUCCESS in 10m 25s\n- gate-nova-python27 http://logs.openstack.org/35/193135/30/check/gate-nova-python27/26ce95a/ : SUCCESS in 15m 55s\n- gate-nova-python34 http://logs.openstack.org/35/193135/30/check/gate-nova-python34/f204e9c/ : SUCCESS in 11m 42s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-full/a1fc758/ : FAILURE in 1h 01m 26s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-postgres-full/fe630de/ : FAILURE in 53m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-neutron-full/b5fe141/ : FAILURE in 1h 21m 50s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/30/check/gate-grenade-dsvm/7e107e1/ : SUCCESS in 47m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-large-ops/afbe9e0/ : SUCCESS in 40m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-neutron-large-ops/c8fb51a/ : SUCCESS in 35m 11s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/30/check/gate-nova-tox-functional/8425195/ : SUCCESS in 15m 49s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/30/check/gate-grenade-dsvm-multinode/2672e49/ : SUCCESS in 1h 01m 02s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9998cd6/ : SUCCESS in 54m 20s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-nova-v20-api/9ca452a/ : SUCCESS in 42m 44s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-nova-v20-api-legacy/fc63afd/ : SUCCESS in 46m 11s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-cells/4705882/ : SUCCESS in 47m 15s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-full-ceph/9a3c166/ : FAILURE in 1h 09m 11s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/c15126b/ : FAILURE in 58m 26s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/30/check/gate-tempest-dsvm-multinode-full/5d94b7e/ : FAILURE in 1h 03m 49s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/30/check/gate-compute-api-guide/2bf3d44//api-guide/build/html/ : SUCCESS in 5m 35s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/30/check/gate-nova-python27-constraints/33405a4/ : SUCCESS in 10m 39s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/30/check/gate-nova-python34-constraints/edcfb5f/ : SUCCESS in 13m 10s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/30/check/gate-nova-docs-constraints/d17f821//doc/build/html/ : SUCCESS in 10m 54s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/30/check/gate-nova-pep8-constraints/0a3399c/ : SUCCESS in 7m 25s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/30/check/gate-nova-tox-constraints-functional-constraints/aea8eba/ : SUCCESS in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"866ee59303032731bffc32dbaeb78bfb79d09836","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 08:10:39.000000000","message":"Patch Set 30:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/30/ : FAILURE in 2h 05m 30s","accounts_in_message":[],"_revision_number":30},{"id":"3745f297fabeb2729452ecc7cf27c96ce0629200","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 08:11:47.000000000","message":"Patch Set 30:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":30},{"id":"15910d5f9306c39d55e835d40fb9f9286ddf9139","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 09:37:14.000000000","message":"Patch Set 30:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/30/ : FAILURE in 1h 24m 02s","accounts_in_message":[],"_revision_number":30},{"id":"226bc9ea65cd05c15dde4b3d69a4586530c1c07a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-15 10:13:17.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/30/Nova-ML2-Sriov : SUCCESS in 57m 37s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":30},{"id":"1d00ce29784be30a6a9d4d5ff2594a0679767f17","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-15 11:53:18.000000000","message":"Patch Set 30: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/30 \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/30 : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"24fd2d7f0a2bd9c1fc9410f51635a49846626836","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-15 12:52:13.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"299e95ec130108a4746b8b97ec6b6943e87d6810","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 13:28:03.000000000","message":"Patch Set 31:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/31/ : FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":31},{"id":"0a7fe22fd9f19d911101ec6afbcbcec5d5486a9e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-15 13:52:12.000000000","message":"Patch Set 31:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/31/check/check-ibm-tempest-dsvm-full/034f9d2/ : FAILURE in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c8633db8b295f59fddcdf3a7a4dbb184d904f8f3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-15 13:58:03.000000000","message":"Patch Set 31: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/31/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/907d7ffad9a2451082280c797d4057b5/index.html : SUCCESS in 16m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/31/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e4b75c06e09c4634803acad11560b966/index.html : SUCCESS in 16m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/31/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f6407bb7347d430385232446d4e5f425/index.html : SUCCESS in 14m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/31/check/real-db-upgrade_nova_percona_user_001:th-percona/1e58feba0fb442569dca3cdedd763f59/index.html : SUCCESS in 16m 07s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/31/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c38f1de68d8f43bdb5f6270799cb62b6/index.html : SUCCESS in 15m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/31/check/real-db-upgrade_nova_percona_user_002:th-percona/e680c5e6a43740b396d9972aca9d5dab/index.html : SUCCESS in 15m 16s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":31},{"id":"9fd6ee89daeada69d9dc43b68392b1ab0dae53ee","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-15 13:59:45.000000000","message":"Patch Set 31: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/31/19702/index.html : SUCCESS in 01 h 03 m 23 s","accounts_in_message":[],"_revision_number":31},{"id":"f6caab170517e5da47a0bd9514ce7735d1b0e474","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-15 14:13:34.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/31/check/dsvm-tempest-xen/349a7ad : FAILURE in 1h 16m 25s","accounts_in_message":[],"_revision_number":31},{"id":"b2e67114a15357489aff1143863262dcba0e39e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 14:46:06.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/31/check/gate-nova-pep8/bcd2eca/ : SUCCESS in 10m 32s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/31/check/gate-nova-docs/046172c//doc/build/html/ : SUCCESS in 16m 07s\n- gate-nova-python27 http://logs.openstack.org/35/193135/31/check/gate-nova-python27/3902145/ : SUCCESS in 15m 28s\n- gate-nova-python34 http://logs.openstack.org/35/193135/31/check/gate-nova-python34/5fb585d/ : SUCCESS in 13m 56s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-full/319babe/ : FAILURE in 1h 00m 00s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-postgres-full/e1f7716/ : FAILURE in 48m 15s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-neutron-full/378ef15/ : FAILURE in 1h 29m 06s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/31/check/gate-grenade-dsvm/8220fcd/ : SUCCESS in 55m 11s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-large-ops/53371d6/ : SUCCESS in 41m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-neutron-large-ops/0d365d4/ : SUCCESS in 32m 58s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/31/check/gate-nova-tox-functional/e64beb4/ : SUCCESS in 17m 47s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/31/check/gate-grenade-dsvm-multinode/981d2aa/ : FAILURE in 49m 35s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/669d68a/ : FAILURE in 43m 03s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-nova-v20-api/4ce1d5f/ : SUCCESS in 47m 52s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-nova-v20-api-legacy/5bed3a4/ : SUCCESS in 46m 58s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-cells/e2a2f79/ : SUCCESS in 48m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-full-ceph/10f108a/ : FAILURE in 1h 01m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/c6b6c61/ : FAILURE in 57m 19s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/31/check/gate-tempest-dsvm-multinode-full/53f5b44/ : FAILURE in 1h 04m 17s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/31/check/gate-compute-api-guide/485efd6//api-guide/build/html/ : SUCCESS in 9m 29s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/31/check/gate-nova-python27-constraints/743089c/ : SUCCESS in 15m 25s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/31/check/gate-nova-python34-constraints/2b6a60a/ : SUCCESS in 14m 01s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/31/check/gate-nova-docs-constraints/5068454//doc/build/html/ : SUCCESS in 14m 54s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/31/check/gate-nova-pep8-constraints/7ba39aa/ : SUCCESS in 9m 29s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/31/check/gate-nova-tox-constraints-functional-constraints/3582b57/ : SUCCESS in 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"53b4331f0ebca00f935f5e5a078959bdd4231c65","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-15 15:44:35.000000000","message":"Patch Set 31:\n\n* pci-test http://52.27.155.124/pci/193135/31 : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"9315f5ef83fbc58b0d715a31f46682d6061e4228","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-15 20:58:00.000000000","message":"Patch Set 31: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/31 \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/193135/31 : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"6124890f7a785b5760945a40e93c3e6c3cf1624d","author":{"_account_id":7173,"name":"Scott DAngelo","email":"scott.dangelo@gmail.com","username":"scottda"},"date":"2015-12-15 22:46:28.000000000","message":"Patch Set 31: Code-Review-1\n\n(2 comments)\n\ncomments inline","accounts_in_message":[],"_revision_number":31},{"id":"8b0c68e5341da948f6abe557e999628bca2b1af1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-15 23:53:22.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/31/Nova-ML2-Sriov : SUCCESS in 55m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":31},{"id":"1aeac0fa55ef7fc9b693f64ba40c3fbfcf611b52","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-16 13:06:32.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"947b2159f16b85dbbeffe722cdca88223c9f1836","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-16 13:10:04.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"e4b8586873c83ebb2a1b58293ad0a5150f8b9e09","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-16 14:15:01.000000000","message":"Patch Set 33: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/33/19988/index.html : SUCCESS in 01 h 01 m 31 s","accounts_in_message":[],"_revision_number":33},{"id":"8df5701f96a71b00ff1c7a24299c773a3a7bf551","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-16 15:00:14.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/33/check/gate-nova-pep8/b94d161/ : SUCCESS in 6m 36s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/33/check/gate-nova-docs/2e4e33a//doc/build/html/ : SUCCESS in 11m 58s\n- gate-nova-python27 http://logs.openstack.org/35/193135/33/check/gate-nova-python27/e9de475/ : FAILURE in 14m 10s\n- gate-nova-python34 http://logs.openstack.org/35/193135/33/check/gate-nova-python34/5f077a2/ : SUCCESS in 10m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-full/78b7ab5/ : FAILURE in 58m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-postgres-full/ea1e498/ : FAILURE in 1h 00m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-neutron-full/2a797cc/ : FAILURE in 1h 23m 35s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/33/check/gate-grenade-dsvm/ee596d9/ : SUCCESS in 47m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-large-ops/8b9879b/ : SUCCESS in 30m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-neutron-large-ops/83631e1/ : SUCCESS in 27m 25s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/33/check/gate-nova-tox-functional/73e4568/ : SUCCESS in 12m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/33/check/gate-grenade-dsvm-multinode/d54c816/ : SUCCESS in 50m 13s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6b2166b/ : FAILURE in 41m 18s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-nova-v20-api/c997248/ : SUCCESS in 53m 39s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-nova-v20-api-legacy/890a0be/ : SUCCESS in 51m 54s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-cells/971dff3/ : SUCCESS in 56m 21s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-full-ceph/ea47923/ : FAILURE in 57m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/69cac4d/ : FAILURE in 1h 05m 38s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/33/check/gate-tempest-dsvm-multinode-full/e83d669/ : FAILURE in 1h 13m 02s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/33/check/gate-compute-api-guide/a9fd3b6//api-guide/build/html/ : SUCCESS in 6m 47s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/33/check/gate-nova-python27-constraints/6e29dbb/ : SUCCESS in 10m 51s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/33/check/gate-nova-python34-constraints/491fbab/ : SUCCESS in 10m 29s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/33/check/gate-nova-docs-constraints/ae1c6f3//doc/build/html/ : SUCCESS in 16m 42s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/33/check/gate-nova-pep8-constraints/a5276fb/ : SUCCESS in 7m 39s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/33/check/gate-nova-tox-constraints-functional-constraints/aec561f/ : SUCCESS in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"ff4108fc58ad1db9d8dffe4de947a9f633f343f5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-16 15:10:57.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/33/ : SUCCESS in 1h 47m 04s","accounts_in_message":[],"_revision_number":33},{"id":"f0c7155ae618090a6ff354727e734b204b999729","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-16 15:34:01.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"fd5d9cf21cae7982bd030fd17fd9a52f9aba3f4e","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-16 15:35:59.000000000","message":"Patch Set 31:\n\n(1 comment)\n\nHi Scott,\n\nThanks for review, please see my comment in line. I also updated the commit message.\n\nThanks and Best Regards,\nIldiko","accounts_in_message":[],"_revision_number":31},{"id":"3d96c91d5855551aae5721bca82adc1c12440754","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-16 16:13:26.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"4fca30fa7d469bb3317aa8f4254d57ae256c0c6c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-16 21:10:17.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://logs.virtuozzoci.qa.sw.ru/35/193135/35/check/check-dsvm-tempest-vz7-exe-minimal/5090c25 : SUCCESS in 57m 36s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"9e91d6583cc0f46066b0617051618cc382af4de2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-16 23:17:10.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/35/check/gate-nova-pep8/68d4284/ : SUCCESS in 8m 57s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/35/check/gate-nova-docs/5ba7e38//doc/build/html/ : SUCCESS in 13m 23s\n- gate-nova-python27 http://logs.openstack.org/35/193135/35/check/gate-nova-python27/dcbb9ec/ : SUCCESS in 14m 00s\n- gate-nova-python34 http://logs.openstack.org/35/193135/35/check/gate-nova-python34/29368a4/ : SUCCESS in 13m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-full/cd15fae/ : FAILURE in 1h 08m 54s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-postgres-full/b6de093/ : FAILURE in 51m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-neutron-full/4da6bb5/ : FAILURE in 1h 29m 35s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/35/check/gate-grenade-dsvm/849e0b8/ : SUCCESS in 45m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-large-ops/c6d38a0/ : SUCCESS in 28m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-neutron-large-ops/eeae129/ : SUCCESS in 29m 42s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/35/check/gate-nova-tox-functional/dd76dbe/ : SUCCESS in 16m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/35/check/gate-grenade-dsvm-multinode/171aed9/ : NOT_REGISTERED\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/361babc/ : FAILURE in 50m 13s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-nova-v20-api/b599f57/ : SUCCESS in 54m 34s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-nova-v20-api-legacy/6b21de8/ : SUCCESS in 36m 26s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-cells/eba57a0/ : SUCCESS in 46m 46s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-full-ceph/f812794/ : FAILURE in 58m 32s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/30678ca/ : FAILURE in 55m 30s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/35/check/gate-tempest-dsvm-multinode-full/413142d/ : NOT_REGISTERED (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/35/check/gate-compute-api-guide/87c49d3//api-guide/build/html/ : SUCCESS in 8m 27s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/35/check/gate-nova-python27-constraints/0748c07/ : SUCCESS in 14m 14s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/35/check/gate-nova-python34-constraints/b1164db/ : SUCCESS in 12m 40s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/35/check/gate-nova-docs-constraints/3e13bdf//doc/build/html/ : SUCCESS in 13m 49s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/35/check/gate-nova-pep8-constraints/98ebc83/ : SUCCESS in 9m 19s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/35/check/gate-nova-tox-constraints-functional-constraints/fc397be/ : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"c2f62ff6977fc6eac155f22b6fd2ba94d44f3178","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-17 05:54:55.000000000","message":"Patch Set 32: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/32 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/54/ : SUCCESS","accounts_in_message":[],"_revision_number":32},{"id":"edad149ed263eee037329b4b334aeec7a67f7ff6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-17 06:52:05.000000000","message":"Patch Set 33:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/33 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/58/ : FAILURE","accounts_in_message":[],"_revision_number":33},{"id":"80170a2354e61c34bbc9846ef80601a81a367441","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-17 06:55:46.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"73e2b14d44708cc1feaec3968a2ddca36c22bdbb","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-17 07:06:02.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/36/check/dsvm-tempest-xen/8cb6733 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":36},{"id":"8700f1d2c2a8a6111d32660f5fbbe383d0a0c0c3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-17 07:25:33.000000000","message":"Patch Set 35: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/35/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/70923143c80d4a38b5ac2dcaa98f31d8/index.html : SUCCESS in 14m 08s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/35/check/real-db-upgrade_nova_mysql_user_001:th-mysql/55fd1cb344da4871b2e8caa554e36929/index.html : SUCCESS in 16m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/35/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2edb7474667b456a9b21ee9783e05136/index.html : SUCCESS in 15m 27s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/35/check/real-db-upgrade_nova_percona_user_001:th-percona/c95ea54d2a7146208b456cb3fea21af1/index.html : SUCCESS in 16m 42s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/35/check/real-db-upgrade_nova_mysql_user_002:th-mysql/77fd9fb3a90e4ff393a408ddcb57381e/index.html : SUCCESS in 16m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/35/check/real-db-upgrade_nova_percona_user_002:th-percona/78280ce6966740038af0bbed9bd730ed/index.html : SUCCESS in 15m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":35},{"id":"3a902a63ff048d19ffe0b97933c95a8718b56df5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-17 07:51:52.000000000","message":"Patch Set 36:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/36/check/check-ibm-tempest-dsvm-full/5510902/ : FAILURE in 54m 52s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"27999b668a936d3cbf584cabe82e65fc9379a9c8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-17 07:57:28.000000000","message":"Patch Set 32:\n\n* pci-test http://52.27.155.124/pci/193135/32 : SUCCESS","accounts_in_message":[],"_revision_number":32},{"id":"0ee355d1a573d7163a29dd9f0c45e8c21c563e69","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-17 08:05:54.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/193135/36/check/dsvm-tempest-kvm/4923e0e : SUCCESS in 1h 03m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":36},{"id":"019367173ce1a1d769b24c5936cf7749782ed85b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-17 08:15:28.000000000","message":"Patch Set 36: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/36/20088/index.html : SUCCESS in 01 h 12 m 09 s","accounts_in_message":[],"_revision_number":36},{"id":"c53cb62ceb16f8e99c2b3bec6f48447521efe24d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-17 08:30:13.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/36/check/gate-nova-pep8/391d02b/ : SUCCESS in 9m 04s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/36/check/gate-nova-docs/3bcd6ee//doc/build/html/ : SUCCESS in 14m 04s\n- gate-nova-python27 http://logs.openstack.org/35/193135/36/check/gate-nova-python27/12d2ff5/ : SUCCESS in 14m 21s\n- gate-nova-python34 http://logs.openstack.org/35/193135/36/check/gate-nova-python34/9a6e917/ : SUCCESS in 13m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-full/ba6486b/ : FAILURE in 1h 02m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-postgres-full/69ce2d6/ : FAILURE in 43m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-neutron-full/ee0ce86/ : FAILURE in 1h 19m 09s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/36/check/gate-grenade-dsvm/2762574/ : SUCCESS in 51m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-large-ops/bbcba4f/ : SUCCESS in 29m 45s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-neutron-large-ops/df0c5b8/ : SUCCESS in 30m 00s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/36/check/gate-nova-tox-functional/122fc12/ : SUCCESS in 17m 10s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/36/check/gate-grenade-dsvm-multinode/bfda85f/ : SUCCESS in 1h 07m 23s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e9496a7/ : SUCCESS in 57m 54s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-nova-v20-api/85d67af/ : SUCCESS in 42m 16s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-nova-v20-api-legacy/29213e3/ : SUCCESS in 43m 16s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-cells/f7562bc/ : SUCCESS in 43m 05s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-full-ceph/7d2624a/ : FAILURE in 1h 03m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/2731ace/ : FAILURE in 50m 27s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/36/check/gate-tempest-dsvm-multinode-full/0bec747/ : FAILURE in 54m 40s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/36/check/gate-compute-api-guide/5f0627b//api-guide/build/html/ : SUCCESS in 8m 36s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/36/check/gate-nova-commit-message/630c6e6/ : FAILURE in 2m 09s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/36/check/gate-nova-python27-constraints/442ffdb/ : SUCCESS in 14m 23s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/36/check/gate-nova-python34-constraints/aa5ea17/ : SUCCESS in 9m 46s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/36/check/gate-nova-docs-constraints/7d4c691//doc/build/html/ : SUCCESS in 14m 05s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/36/check/gate-nova-pep8-constraints/ceb2f5b/ : SUCCESS in 9m 19s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/36/check/gate-nova-tox-constraints-functional-constraints/5c2c0db/ : SUCCESS in 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"27d05b79c13deb80db4151d759bab1b34a7e70b6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-17 08:47:42.000000000","message":"Patch Set 33:\n\n* pci-test http://52.27.155.124/pci/193135/33 : SUCCESS","accounts_in_message":[],"_revision_number":33},{"id":"6395fbac9e641249c471d5bda7e62b495dcf255e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-17 10:11:51.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/36/ : SUCCESS in 2h 38m 59s","accounts_in_message":[],"_revision_number":36},{"id":"978beac277593cdb006cad3bc5a287cf12518252","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-17 10:42:03.000000000","message":"Patch Set 34:\n\n* pci-test http://52.27.155.124/pci/193135/34 : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"fbf84d65773b3234a65ba8ca671e69213d88feff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-17 10:42:06.000000000","message":"Patch Set 34:\n\nBuild failed. Logs: http://208.91.1.172/logs/193135/34 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/69/ : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"eda547124843a62f7515c99794028251b2842b2c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-17 11:17:33.000000000","message":"Patch Set 36: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/36/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1f4b99966aa0415aa85dc7890ef40739/index.html : SUCCESS in 15m 11s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/36/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d0b9db77e1494d08829f4e15979daaf5/index.html : SUCCESS in 16m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/36/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ad17b8a8028143438532eb05209aca26/index.html : SUCCESS in 13m 22s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/36/check/real-db-upgrade_nova_percona_user_001:th-percona/7922ccc83886438ba8dba3a5aabb247d/index.html : SUCCESS in 16m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/36/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5a9b68084aa64240a11d999ac99d6f32/index.html : SUCCESS in 14m 58s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/36/check/real-db-upgrade_nova_percona_user_002:th-percona/77e7be4abfc547409299aef05b5c7a9e/index.html : SUCCESS in 15m 16s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":36},{"id":"fc3e387bfe61bddfd88a7b68ebcc6c9aa1797948","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-17 12:23:30.000000000","message":"Patch Set 35:\n\n* pci-test http://52.27.155.124/pci/193135/35 : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"d4e53fecb7c84f9560960a6dd0c9376920c220e3","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-12-17 12:25:50.000000000","message":"Patch Set 36: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":36},{"id":"9a6efaececc8a95aa2fd15e2d0286c1411ee4e96","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-17 15:21:16.000000000","message":"Patch Set 35: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/35 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/81/ : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"5eeece400036ea27bc555bbda491afe7e92b82ce","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-17 19:11:13.000000000","message":"Patch Set 36:\n\n* pci-test http://52.27.155.124/pci/193135/36 : FAILURE","accounts_in_message":[],"_revision_number":36},{"id":"a77530d0024cf1dd744dc5285c31f5c1ed174a20","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2015-12-17 20:13:48.000000000","message":"Patch Set 36:\n\n(2 comments)","accounts_in_message":[],"_revision_number":36},{"id":"cee5f3b20576e5811a924c31ad518739f0bcecd0","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-17 21:10:38.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"3619f104f30ae5b153f3a25dd144988f30ee4edd","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-17 21:11:39.000000000","message":"Patch Set 37: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":37},{"id":"93e71b8be0f6631136037920bc5ff8ed806a3aea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-17 22:05:15.000000000","message":"Patch Set 37:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/37/check/check-ibm-tempest-dsvm-full/01e0dce/ : FAILURE in 53m 18s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"a4cd5cbd060ce873d3a2f8c5719c0bde1443d235","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-17 22:20:00.000000000","message":"Patch Set 37: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/37/20299/index.html : SUCCESS in 01 h 04 m 09 s","accounts_in_message":[],"_revision_number":37},{"id":"77b5ec1cda4f81d942e14572caa9192e07c52a51","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-18 01:17:45.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/37/check/gate-nova-pep8/49831a8/ : SUCCESS in 8m 42s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/37/check/gate-nova-docs/ebe2f53//doc/build/html/ : SUCCESS in 14m 36s\n- gate-nova-python27 http://logs.openstack.org/35/193135/37/check/gate-nova-python27/15720a6/ : FAILURE in 14m 25s\n- gate-nova-python34 http://logs.openstack.org/35/193135/37/check/gate-nova-python34/aac3fcb/ : SUCCESS in 10m 43s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-full/66180e1/ : FAILURE in 56m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-postgres-full/3ec38d4/ : FAILURE in 42m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-neutron-full/4b16713/ : FAILURE in 1h 22m 39s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/37/check/gate-grenade-dsvm/d99795f/ : SUCCESS in 50m 42s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-large-ops/c6bd94d/ : SUCCESS in 45m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-neutron-large-ops/22e260b/ : SUCCESS in 27m 49s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/37/check/gate-nova-tox-functional/18abf24/ : SUCCESS in 14m 54s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/37/check/gate-grenade-dsvm-multinode/56b9990/ : FAILURE in 8m 04s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/fd970aa/ : SUCCESS in 1h 05m 51s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-nova-v20-api/b34b6d6/ : SUCCESS in 47m 42s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-nova-v20-api-legacy/0869f51/ : SUCCESS in 51m 11s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-cells/9127ed0/ : SUCCESS in 34m 50s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-full-ceph/bccde47/ : FAILURE in 46m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/0bfe1e1/ : FAILURE in 49m 27s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/37/check/gate-tempest-dsvm-multinode-full/eb485a5/ : FAILURE in 47m 52s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/37/check/gate-compute-api-guide/8ac70b2//api-guide/build/html/ : SUCCESS in 6m 49s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/37/check/gate-nova-commit-message/927c8b0/ : FAILURE in 1m 54s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/37/check/gate-nova-python27-constraints/ba3aee7/ : FAILURE in 14m 33s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/37/check/gate-nova-python34-constraints/9128248/ : SUCCESS in 13m 16s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/37/check/gate-nova-docs-constraints/9959143//doc/build/html/ : SUCCESS in 14m 12s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/37/check/gate-nova-pep8-constraints/d986507/ : SUCCESS in 8m 25s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/37/check/gate-nova-tox-constraints-functional-constraints/7949cbc/ : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"056cfbd588c3d80323e0004b95c3c2458d0a4827","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-18 01:24:57.000000000","message":"Patch Set 37:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/37/ : FAILURE in 1h 56m 57s","accounts_in_message":[],"_revision_number":37},{"id":"c560c861c8716e3a4c1e014b023f97683cd4b15a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-18 06:59:47.000000000","message":"Patch Set 37: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/37/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0b4844fddd314f8881c11f4ee858f8a0/index.html : SUCCESS in 18m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/37/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a36d889e124a44ec86e97b2d06231974/index.html : SUCCESS in 17m 48s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/37/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8a19605e10974450ab30e9a71a68585b/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/37/check/real-db-upgrade_nova_percona_user_001:th-percona/4d14a656223a460789e86f2ca7877368/index.html : SUCCESS in 15m 37s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/37/check/real-db-upgrade_nova_mysql_user_002:th-mysql/117b29c3caa3457c92162da106b9dde8/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/37/check/real-db-upgrade_nova_percona_user_002:th-percona/3be989b6be7f4dcd86691d757dc35fab/index.html : SUCCESS in 12m 35s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":37},{"id":"5b81fe5942fa5573a477a825ed1152b4f2c096d3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-18 08:52:55.000000000","message":"Patch Set 37:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/193135/37/check/dsvm-tempest-kvm/2cb6469 : SUCCESS in 44m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":37},{"id":"7eba0941323cf3f05af158dbf335b98560d70192","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-18 10:23:07.000000000","message":"Patch Set 36: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/36 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/127/ : SUCCESS","accounts_in_message":[],"_revision_number":36},{"id":"7ac3f66b9a6eead9c705c373384ed4473d9bfcc7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-18 13:53:09.000000000","message":"Patch Set 37:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/37/Nova-ML2-Sriov : SUCCESS in 57m 00s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":37},{"id":"fdf3766ab34b4be1b087d80100fbb544ede93b33","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-18 14:31:45.000000000","message":"Patch Set 37:\n\n* pci-test http://52.27.155.124/pci/193135/37 : SUCCESS","accounts_in_message":[],"_revision_number":37},{"id":"3a0029158041af513a3220487200fa2f86014a9b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-21 03:51:38.000000000","message":"Patch Set 37: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/37 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/297/ : SUCCESS","accounts_in_message":[],"_revision_number":37},{"id":"32b392451a63a300be05f7466f722ed15d2e3d71","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-23 05:57:54.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"bd8542297fcf83240b4a6971a291024390f8189d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-23 06:16:00.000000000","message":"Patch Set 38:\n\n* pci-test http://52.27.155.124/pci/193135/38 : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"71d02b870a70d974ce332f29e03abe098cca0599","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-23 06:20:43.000000000","message":"Patch Set 38: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/38/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9252f03f07714201926bcb9444e75eaa/index.html : SUCCESS in 11m 55s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/38/check/real-db-upgrade_nova_mysql_user_001:th-mysql/50dc1e8bb9d7437ea3c0cb073c755710/index.html : SUCCESS in 16m 10s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/38/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b151cc7df1e949e7ac060d60aeca0119/index.html : SUCCESS in 15m 09s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/38/check/real-db-upgrade_nova_percona_user_001:th-percona/d8188db76c4f4960a35ccb214ab8f2f1/index.html : SUCCESS in 16m 15s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/38/check/real-db-upgrade_nova_mysql_user_002:th-mysql/99612c6dbec54397a21a7e741bd12cba/index.html : SUCCESS in 12m 53s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/38/check/real-db-upgrade_nova_percona_user_002:th-percona/c39d94cc15064128a947442a94dd95b5/index.html : SUCCESS in 15m 07s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":38},{"id":"235bffcee4bee73e66c9fd97b0652fe169e3229a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-23 06:21:57.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/35/193135/38/check/dsvm-tempest-kvm/7f7d5b5 : FAILURE in 22m 20s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":38},{"id":"df59f03bd5950291d1a1e07f95d7909e3e5f916b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-23 06:50:21.000000000","message":"Patch Set 38:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/38/check/check-ibm-tempest-dsvm-full/cc1354f/ : FAILURE in 51m 11s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"bc870d64b3bcf7c76e245cb8680858e56cc43f17","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-23 06:57:56.000000000","message":"Patch Set 38: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/38/20783/index.html : SUCCESS in 56 m 04 s","accounts_in_message":[],"_revision_number":38},{"id":"b43092fe61a2660725a1ec69fca2f377ea2c9ff5","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-23 07:06:33.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/38/check/dsvm-tempest-xen/23d570f : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":38},{"id":"038c56184f6160d502355cd17d8229615361b4ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-23 07:09:20.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/38/check/gate-nova-pep8/6a6e441/ : SUCCESS in 9m 26s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/38/check/gate-nova-docs/7e08411//doc/build/html/ : SUCCESS in 11m 30s\n- gate-nova-python27 http://logs.openstack.org/35/193135/38/check/gate-nova-python27/ee90f44/ : FAILURE in 9m 32s\n- gate-nova-python34 http://logs.openstack.org/35/193135/38/check/gate-nova-python34/879e3a9/ : SUCCESS in 13m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-full/7d49d32/ : FAILURE in 55m 13s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-postgres-full/140e22e/ : FAILURE in 52m 14s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-neutron-full/7e28ded/ : FAILURE in 1h 08m 13s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/38/check/gate-grenade-dsvm/a5bc87c/ : SUCCESS in 51m 01s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-large-ops/f35aabf/ : SUCCESS in 37m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-neutron-large-ops/1a2cbe4/ : SUCCESS in 36m 21s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/38/check/gate-nova-tox-functional/d0b462b/ : SUCCESS in 11m 33s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/38/check/gate-grenade-dsvm-multinode/2b01982/ : SUCCESS in 47m 24s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/1c0c9fc/ : SUCCESS in 52m 59s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-nova-v20-api/6bb6bfc/ : SUCCESS in 55m 12s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-nova-v20-api-legacy/83db953/ : SUCCESS in 44m 01s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-cells/85bff7b/ : SUCCESS in 52m 15s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-full-ceph/8dde8da/ : FAILURE in 59m 16s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/48ab99a/ : FAILURE in 57m 06s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/38/check/gate-tempest-dsvm-multinode-full/2fd2bb9/ : FAILURE in 50m 25s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/38/check/gate-compute-api-guide/0da8a92//api-guide/build/html/ : SUCCESS in 6m 34s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/38/check/gate-nova-commit-message/ceaeae7/ : SUCCESS in 1m 14s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/38/check/gate-nova-python27-constraints/5c4c101/ : FAILURE in 15m 07s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/38/check/gate-nova-python34-constraints/dfefe31/ : SUCCESS in 13m 06s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/38/check/gate-nova-docs-constraints/5dca30d//doc/build/html/ : SUCCESS in 14m 56s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/38/check/gate-nova-pep8-constraints/8038e58/ : SUCCESS in 9m 48s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/38/check/gate-nova-tox-constraints-functional-constraints/921811f/ : SUCCESS in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"12f768ae0e0d904118fe0a85e5ddbc6e3cc990e1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-23 07:09:48.000000000","message":"Patch Set 38:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/38/ : FAILURE in 1h 10m 33s","accounts_in_message":[],"_revision_number":38},{"id":"aee5dd25f48dbe3db62ac7ff24414355e494b10b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-23 11:29:20.000000000","message":"Patch Set 38:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/38/check/check-dsvm-tempest-vz7-exe-minimal/7a767bc : FAILURE in 3h 05m 09s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":38},{"id":"69fc952fd225c0c314b6796e5af85fcb3421a488","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-23 12:23:34.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/38/Nova-ML2-Sriov : SUCCESS in 57m 49s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":38},{"id":"cba73cf34ec09e6ed44fcfa842550c8834f89565","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-24 12:51:08.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"67f095d64d55ffe025b332d78432b7f7f4f7ef1d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-24 13:20:12.000000000","message":"Patch Set 39:\n\n* pci-test http://52.27.155.124/pci/193135/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"41c665e38630f22071dde78da04059b0202df27c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-24 13:40:07.000000000","message":"Patch Set 39: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/39/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b4047b46e2744d9b9ce2bede01ee4440/index.html : SUCCESS in 14m 45s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/39/check/real-db-upgrade_nova_mysql_user_001:th-mysql/31afb704824c47acad9c48a86686d72a/index.html : SUCCESS in 16m 06s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/39/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/50949eb8f27b48ba9ea9a701b4997b93/index.html : SUCCESS in 13m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/39/check/real-db-upgrade_nova_percona_user_001:th-percona/24e1b68846d04871b20b9735e00314c1/index.html : SUCCESS in 16m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/39/check/real-db-upgrade_nova_mysql_user_002:th-mysql/83f5088cb0df4848bad699409514aead/index.html : SUCCESS in 15m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/39/check/real-db-upgrade_nova_percona_user_002:th-percona/0e5a46d824784341b044f98fa44e4f61/index.html : SUCCESS in 14m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":39},{"id":"b594c307b6788bb511b23be84f447598362cc91a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-24 13:43:03.000000000","message":"Patch Set 39:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/39/check/check-ibm-tempest-dsvm-full/3f106e4/ : FAILURE in 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"730f6965b6754eeb048e8581b8ac03599789bf8d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-24 13:46:02.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/39/check/dsvm-tempest-xen/029182d : FAILURE in 24m 30s","accounts_in_message":[],"_revision_number":39},{"id":"10595c37ff760319a25b78b58578503ac43e85f9","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-24 13:51:49.000000000","message":"Patch Set 39:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/39/check/check-dsvm-tempest-vz7-exe-minimal/1b8f1d4 : FAILURE in 9m 06s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":39},{"id":"1c79e918b3fbd160c995296755d732392724788f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-24 13:55:28.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/39/ : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":39},{"id":"39d1dfefb1c3ca57ece0cf64b08970c6cfc416aa","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-24 13:59:19.000000000","message":"Patch Set 39: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/39/20935/index.html : SUCCESS in 01 h 02 m 36 s","accounts_in_message":[],"_revision_number":39},{"id":"0da965e762e01e68555d7ca42d3ae6d56bf5406b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-24 14:01:40.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/39/check/gate-nova-pep8/919a061/ : SUCCESS in 10m 07s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/39/check/gate-nova-docs/a364b4c//doc/build/html/ : SUCCESS in 15m 13s\n- gate-nova-python27 http://logs.openstack.org/35/193135/39/check/gate-nova-python27/43ceb73/ : SUCCESS in 14m 36s\n- gate-nova-python34 http://logs.openstack.org/35/193135/39/check/gate-nova-python34/d9c6ed9/ : SUCCESS in 13m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-full/86081cd/ : FAILURE in 59m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-postgres-full/97b93d1/ : FAILURE in 48m 17s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-neutron-full/40b7c81/ : FAILURE in 1h 09m 14s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/39/check/gate-grenade-dsvm/3e96693/ : SUCCESS in 49m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-large-ops/e9204a9/ : SUCCESS in 28m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-neutron-large-ops/fa9609a/ : SUCCESS in 29m 16s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/39/check/gate-nova-tox-functional/6b8cd8e/ : SUCCESS in 12m 50s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/39/check/gate-grenade-dsvm-multinode/6971f17/ : SUCCESS in 48m 49s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b6b6453/ : SUCCESS in 48m 41s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-nova-v20-api/2be0196/ : SUCCESS in 35m 10s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-nova-v20-api-legacy/ef0a5bd/ : SUCCESS in 32m 24s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-cells/6233d46/ : SUCCESS in 45m 36s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-full-ceph/6d1ebc6/ : FAILURE in 50m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/65ae22b/ : FAILURE in 53m 16s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/39/check/gate-tempest-dsvm-multinode-full/e33b6c8/ : FAILURE in 1h 04m 13s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/39/check/gate-compute-api-guide/684ac16//api-guide/build/html/ : SUCCESS in 8m 54s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/39/check/gate-nova-commit-message/9f274e9/ : SUCCESS in 1m 29s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/39/check/gate-nova-python27-constraints/d7b8772/ : SUCCESS in 14m 31s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/39/check/gate-nova-python34-constraints/8be293d/ : SUCCESS in 10m 31s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/39/check/gate-nova-docs-constraints/61d1ec6//doc/build/html/ : SUCCESS in 11m 39s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/39/check/gate-nova-pep8-constraints/bca79e4/ : SUCCESS in 9m 13s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/39/check/gate-nova-tox-constraints-functional-constraints/94fe8ea/ : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"77d6d0fe54ca481babbee16383be07dd9b4f205f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-24 14:20:02.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/39/Nova-ML2-Sriov : FAILURE in 2m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":39},{"id":"976cf28345e9733bfe352ef69105320c79ad6238","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-24 19:28:54.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/39/Nova-ML2-Sriov : SUCCESS in 57m 26s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":39},{"id":"92db6f53dbddaf67c48b1827f6f03d95e2616276","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-25 03:06:07.000000000","message":"Patch Set 39:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/39/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":39},{"id":"5a469f70251505a8473bb326e4866214ac45bd42","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-25 10:38:50.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"9a92aa13ccfda73c8d44340be3de2b02fe121caf","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-25 11:03:54.000000000","message":"Patch Set 40: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/40/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0fbbd9ea69414123887a47e69c950870/index.html : SUCCESS in 14m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/40/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6a8c5ce691174e73895a594f6a80bcd2/index.html : SUCCESS in 16m 52s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/40/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b27f7e3185544890995ca0b81f601649/index.html : SUCCESS in 13m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/40/check/real-db-upgrade_nova_percona_user_001:th-percona/442e5ff3206041b086d17634bf5ce277/index.html : SUCCESS in 13m 09s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/40/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ed7f4c12edd147668a0f233504e703f1/index.html : SUCCESS in 13m 02s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/40/check/real-db-upgrade_nova_percona_user_002:th-percona/6ef637876b0745e2bfa1c8f65c3d1b7f/index.html : SUCCESS in 12m 41s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":40},{"id":"d5abfeebb479b8c21a8a268f2a322d865c29f671","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-25 11:11:12.000000000","message":"Patch Set 40:\n\n* pci-test http://52.27.155.124/pci/193135/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"108737fddacb5938726a36bd01f0031c34134ee6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-25 11:28:00.000000000","message":"Patch Set 40:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/40/check/check-ibm-tempest-dsvm-full/6ce5b86/ : SUCCESS in 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"134925c1ddc40657771ee67af1f92ad12ed70b28","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-25 11:40:07.000000000","message":"Patch Set 40: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/40/21001/index.html : SUCCESS in 57 m 02 s","accounts_in_message":[],"_revision_number":40},{"id":"d7acf67f2631b1f7d0816415fd49a893792a4473","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-25 11:42:17.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/40/ : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":40},{"id":"100dc98ae64016c5256fe32da660ba3d4bb55552","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-25 11:46:30.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/40/check/gate-nova-pep8/55b041e/ : SUCCESS in 9m 31s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/40/check/gate-nova-docs/11972e0//doc/build/html/ : SUCCESS in 14m 14s\n- gate-nova-python27 http://logs.openstack.org/35/193135/40/check/gate-nova-python27/4b70a1b/ : SUCCESS in 11m 42s\n- gate-nova-python34 http://logs.openstack.org/35/193135/40/check/gate-nova-python34/5e48069/ : SUCCESS in 11m 08s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-full/215211b/ : SUCCESS in 44m 03s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-postgres-full/96e8d94/ : SUCCESS in 44m 07s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-neutron-full/86e7e36/ : SUCCESS in 1h 06m 18s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/40/check/gate-grenade-dsvm/c9c9242/ : SUCCESS in 48m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-large-ops/f25db0a/ : SUCCESS in 27m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-neutron-large-ops/ae07c42/ : SUCCESS in 27m 27s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/40/check/gate-nova-tox-functional/1292d74/ : SUCCESS in 14m 54s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/40/check/gate-grenade-dsvm-multinode/064d803/ : SUCCESS in 49m 15s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/53fa896/ : SUCCESS in 35m 28s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-nova-v20-api/3f2a816/ : SUCCESS in 31m 51s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-nova-v20-api-legacy/7cb99b9/ : SUCCESS in 43m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-cells/1fca62e/ : SUCCESS in 36m 37s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-full-ceph/b7b2c45/ : SUCCESS in 46m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/94510c3/ : SUCCESS in 45m 46s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/40/check/gate-tempest-dsvm-multinode-full/99985aa/ : FAILURE in 58m 15s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/40/check/gate-compute-api-guide/0dc17d0//api-guide/build/html/ : SUCCESS in 9m 31s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/40/check/gate-nova-commit-message/40586e6/ : SUCCESS in 1m 34s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/40/check/gate-nova-python27-constraints/5942b5d/ : SUCCESS in 11m 56s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/40/check/gate-nova-python34-constraints/1924f9c/ : SUCCESS in 13m 29s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/40/check/gate-nova-docs-constraints/b9762da//doc/build/html/ : SUCCESS in 14m 44s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/40/check/gate-nova-pep8-constraints/6a08070/ : SUCCESS in 9m 53s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/40/check/gate-nova-tox-constraints-functional-constraints/dec501a/ : SUCCESS in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"2eb6326a60ec923f33b1eefd31d7eb6c5c4f9099","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-25 11:51:08.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/40/check/dsvm-tempest-xen/5831280 : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":40},{"id":"247e5bd23eed36a9b4d778f1f8f0fb71d5120b55","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-25 12:01:08.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/40/check/check-dsvm-tempest-vz7-exe-minimal/28bb3da : SUCCESS in 43m 56s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":40},{"id":"9706bbc03cfd92620a278b42df356be2d76525e2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-25 13:04:36.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/40/Nova-ML2-Sriov : SUCCESS in 57m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":40},{"id":"68f2fdd8aacaa5b7fcefb7fa8ef87b26e2a9d783","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-25 17:53:02.000000000","message":"Patch Set 40:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/40/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":40},{"id":"34fdd420a62108629cc098e494e6385d519c084c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-25 18:52:10.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.","accounts_in_message":[],"_revision_number":41},{"id":"4393f759bb091cfee3587ac400644ffeda04c412","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-25 19:08:43.000000000","message":"Patch Set 41:\n\n* pci-test http://52.27.155.124/pci/193135/41 : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"01cd140753fc5d2726b823e885e5dfed62f29d4a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-25 19:15:32.000000000","message":"Patch Set 41: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/41/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/34399003a10d4d61a169357701bb2f31/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/41/check/real-db-upgrade_nova_mysql_user_001:th-mysql/46fb932fd0344320899a1d1a272f4f8a/index.html : SUCCESS in 15m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/41/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1ce17f1b7cb448428d47918d4783fe70/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/41/check/real-db-upgrade_nova_percona_user_001:th-percona/60e887aa451e4fd7aa3d047e6a0fafbc/index.html : SUCCESS in 14m 02s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/41/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6f5b745d314b458885a65c8f4965017e/index.html : SUCCESS in 14m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/41/check/real-db-upgrade_nova_percona_user_002:th-percona/03a5104cd3d84656b20bfffcaf430f3e/index.html : SUCCESS in 15m 14s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":41},{"id":"6ba7fe8875c5661d9889f911a764cb00e918a611","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-25 19:36:21.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/41/check/check-dsvm-tempest-vz7-exe-minimal/3b75617 : SUCCESS in 42m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":41},{"id":"949691f2b9d4daacb36e49dde5a6c80ec7d066d4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-25 19:51:03.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/41/Nova-ML2-Sriov : SUCCESS in 56m 31s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":41},{"id":"e403ab034ea10ee3e0ea09a27cb2bdad9b026ce6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-25 19:54:07.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/41/check/gate-nova-pep8/40d9f21/ : SUCCESS in 9m 35s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/41/check/gate-nova-docs/a84aed0//doc/build/html/ : SUCCESS in 14m 54s\n- gate-nova-python27 http://logs.openstack.org/35/193135/41/check/gate-nova-python27/bc012e4/ : SUCCESS in 15m 42s\n- gate-nova-python34 http://logs.openstack.org/35/193135/41/check/gate-nova-python34/1e8a1e6/ : SUCCESS in 10m 18s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-full/7a5438d/ : SUCCESS in 45m 38s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-postgres-full/e9eb758/ : SUCCESS in 49m 15s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-neutron-full/e2f488b/ : SUCCESS in 58m 39s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/41/check/gate-grenade-dsvm/8cf4d38/ : SUCCESS in 40m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-large-ops/2257ab9/ : SUCCESS in 38m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-neutron-large-ops/2936fe0/ : SUCCESS in 38m 03s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/41/check/gate-nova-tox-functional/a6ccd7b/ : SUCCESS in 12m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/41/check/gate-grenade-dsvm-multinode/941651b/ : SUCCESS in 47m 34s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/4b1e8f8/ : SUCCESS in 37m 20s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-nova-v20-api/a5cb6be/ : SUCCESS in 43m 13s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-nova-v20-api-legacy/ae3a98c/ : SUCCESS in 42m 01s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-cells/e21c466/ : SUCCESS in 41m 58s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-full-ceph/45ca2b0/ : SUCCESS in 52m 24s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/1b0767d/ : SUCCESS in 53m 43s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/41/check/gate-tempest-dsvm-multinode-full/adeecd3/ : SUCCESS in 49m 38s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/41/check/gate-compute-api-guide/cd55b0d//api-guide/build/html/ : SUCCESS in 8m 19s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/41/check/gate-nova-commit-message/68f5b8c/ : SUCCESS in 1m 16s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/41/check/gate-nova-python27-constraints/821e26d/ : SUCCESS in 15m 27s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/41/check/gate-nova-python34-constraints/2fa7c79/ : SUCCESS in 10m 41s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/41/check/gate-nova-docs-constraints/e36479f//doc/build/html/ : SUCCESS in 11m 34s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/41/check/gate-nova-pep8-constraints/debeece/ : SUCCESS in 7m 46s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/41/check/gate-nova-tox-constraints-functional-constraints/3a30b3f/ : SUCCESS in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"98abcd48f0b12539e3804936b91fa6088b88dd35","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-25 19:54:45.000000000","message":"Patch Set 41: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/41/21020/index.html : SUCCESS in 59 m 31 s","accounts_in_message":[],"_revision_number":41},{"id":"b35c37de70ab6ab0b0740c87201010509ec0fb2c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-25 19:56:07.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/41/ : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":41},{"id":"b98541093889efa154e415ef518d4d891a05ac69","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-25 20:03:32.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/41/check/dsvm-tempest-xen/570f8c0 : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":41},{"id":"906317cb9c6709e4dd5b72aa380fc6ee755c275a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-25 22:01:40.000000000","message":"Patch Set 41:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/41/check/check-ibm-tempest-dsvm-full/90f0ebc/ : SUCCESS in 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"09845661ce1060d42118832c893ce23b1e7bbbbc","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-26 09:46:22.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"179c468c0b6dfb803e28d49012a67fd04796a6db","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-26 10:04:18.000000000","message":"Patch Set 42:\n\n* pci-test http://52.27.155.124/pci/193135/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"27b4ca9b083b509a1ecce0221228da30c214eeca","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-26 10:11:16.000000000","message":"Patch Set 42: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/42/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bbaa05dd5eb4401e876921f14c09896a/index.html : SUCCESS in 12m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/42/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3627ae6bc8584df991dc59e1ff40bf0c/index.html : SUCCESS in 16m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/42/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/749daadf3e614ec4b82cf89615286478/index.html : SUCCESS in 12m 19s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/42/check/real-db-upgrade_nova_percona_user_001:th-percona/1d6c617e4fa84e9f9a066d18dda6179d/index.html : SUCCESS in 15m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/42/check/real-db-upgrade_nova_mysql_user_002:th-mysql/b73ab77e115d4c6d95aa581101984356/index.html : SUCCESS in 12m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/42/check/real-db-upgrade_nova_percona_user_002:th-percona/12d773ab40eb4947b3bf0b730733751b/index.html : SUCCESS in 12m 33s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":42},{"id":"3b529fd54303bc7834a64dcf5b0a95ef1e820d93","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-26 10:31:10.000000000","message":"Patch Set 42:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/42/check/check-dsvm-tempest-vz7-exe-minimal/51b061e : FAILURE in 41m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":42},{"id":"dddf767989d63d1dbe70800c34134da0dcb9fb63","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-26 10:38:10.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.","accounts_in_message":[],"_revision_number":43},{"id":"1fed423a8a375fc3212636ff5a98a94ed8fabfe8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-26 10:56:15.000000000","message":"Patch Set 43:\n\n* pci-test http://52.27.155.124/pci/193135/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"90dac740fb930545e5facdd857075819b92fed75","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-26 11:02:05.000000000","message":"Patch Set 43: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/43/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/218f41d5e6a14c77918cf7c3da623425/index.html : SUCCESS in 12m 38s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/43/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e04192539d7044c7b9cfec212f75df68/index.html : SUCCESS in 16m 25s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/43/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8bb9d5dfff1b48e28720a3c4364bec76/index.html : SUCCESS in 12m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/43/check/real-db-upgrade_nova_percona_user_001:th-percona/d0e5c7640e94464f9e0cb2baf3944d19/index.html : SUCCESS in 16m 04s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/43/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4a3c5a741bed4b708dc9295b10309887/index.html : SUCCESS in 12m 32s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/43/check/real-db-upgrade_nova_percona_user_002:th-percona/3872263eb2994826a97613d81e20c00f/index.html : SUCCESS in 14m 44s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":43},{"id":"7913d6e4d1e4433a511e79a71e8c1a7ea46e6747","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-26 11:25:04.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/43/check/check-dsvm-tempest-vz7-exe-minimal/1a17b7d : SUCCESS in 44m 18s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":43},{"id":"54ec4fa9d1ec79bbe52c04b225e9938472f5cd23","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-26 11:33:45.000000000","message":"Patch Set 43:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/43/check/check-ibm-tempest-dsvm-full/6ce887b/ : FAILURE in 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"d5e972900f3a5623aa9e3921a9bcd331008fbd4f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-26 11:38:50.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/43/Nova-ML2-Sriov : SUCCESS in 58m 09s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":43},{"id":"5f853d461ade02314b0ac0c034547a109d5dd0e7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-26 11:43:08.000000000","message":"Patch Set 43: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/43/21043/index.html : FAILURE in 01 h 02 m 50 s","accounts_in_message":[],"_revision_number":43},{"id":"9848cb29fd7d163a358cf9d801fdb7f322a14089","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-26 11:48:38.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/43/check/dsvm-tempest-xen/f65f552 : FAILURE in 1h 08m 43s","accounts_in_message":[],"_revision_number":43},{"id":"3a93b01a9c6d733510244f3b3dc5c691b4dfe2f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-26 12:07:26.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/43/check/gate-nova-pep8/a5442e4/ : SUCCESS in 8m 56s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/43/check/gate-nova-docs/744848f//doc/build/html/ : SUCCESS in 14m 36s\n- gate-nova-python27 http://logs.openstack.org/35/193135/43/check/gate-nova-python27/9e04589/ : FAILURE in 14m 18s\n- gate-nova-python34 http://logs.openstack.org/35/193135/43/check/gate-nova-python34/7f771dc/ : SUCCESS in 13m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-full/66a3a75/ : FAILURE in 58m 52s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-postgres-full/6c191f6/ : FAILURE in 1h 02m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-neutron-full/863d5b4/ : FAILURE in 1h 26m 13s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/43/check/gate-grenade-dsvm/8ad4fa7/ : FAILURE in 42m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-large-ops/6d45eb5/ : SUCCESS in 36m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-neutron-large-ops/411f8c9/ : SUCCESS in 31m 22s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/43/check/gate-nova-tox-functional/233b97c/ : SUCCESS in 15m 47s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/43/check/gate-grenade-dsvm-multinode/be174cd/ : FAILURE in 49m 23s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e0b9459/ : SUCCESS in 41m 11s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-nova-v20-api/4c0bd24/ : FAILURE in 46m 52s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-nova-v20-api-legacy/19720ae/ : FAILURE in 50m 16s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-cells/2a5e656/ : FAILURE in 1h 04m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-full-ceph/e9457a1/ : FAILURE in 55m 26s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5954202/ : FAILURE in 1h 06m 15s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/43/check/gate-tempest-dsvm-multinode-full/8e7d6db/ : FAILURE in 1h 16m 56s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/43/check/gate-compute-api-guide/26dbd09//api-guide/build/html/ : SUCCESS in 8m 49s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/43/check/gate-nova-commit-message/39a86f1/ : SUCCESS in 1m 15s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/43/check/gate-nova-python27-constraints/cdf7400/ : FAILURE in 11m 36s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/43/check/gate-nova-python34-constraints/95f75fa/ : SUCCESS in 11m 17s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/43/check/gate-nova-docs-constraints/7aa5051//doc/build/html/ : SUCCESS in 9m 55s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/43/check/gate-nova-pep8-constraints/d76c223/ : SUCCESS in 7m 39s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/43/check/gate-nova-tox-constraints-functional-constraints/3ad0b96/ : SUCCESS in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"0dcfd2e540a7852e7265baae9478a05d0769a443","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-26 13:03:24.000000000","message":"Patch Set 43:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/43/ : FAILURE in 2h 23m 05s","accounts_in_message":[],"_revision_number":43},{"id":"3579cf9734808dff9606e5d55139709f7b654600","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-26 23:03:33.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.","accounts_in_message":[],"_revision_number":44},{"id":"84bc4555a19e6142659a6bf8b43e7037d47fc97a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-26 23:21:15.000000000","message":"Patch Set 44:\n\n* pci-test http://52.27.155.124/pci/193135/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"40bcfd8db4128544c76100189a91779ea0cdea82","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-26 23:28:10.000000000","message":"Patch Set 44: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/44/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/39417df5f2654762ba7ea846755a9503/index.html : SUCCESS in 13m 33s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/44/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3ecaf4d53e614cc7b5d9b8a9dda2ae1d/index.html : SUCCESS in 13m 21s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/44/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ddb45659bac34bb5969790ce33372d8c/index.html : SUCCESS in 14m 10s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/44/check/real-db-upgrade_nova_percona_user_001:th-percona/eb3a4d65505b4d8d9945e95d233a68a6/index.html : SUCCESS in 13m 44s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/44/check/real-db-upgrade_nova_mysql_user_002:th-mysql/baa0fcfdbb00468a90acec8316ecf852/index.html : SUCCESS in 15m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/44/check/real-db-upgrade_nova_percona_user_002:th-percona/7a0b50c5bf37433ebcebaa7813079b38/index.html : SUCCESS in 14m 54s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":44},{"id":"1347c5946446d12511bb221477d3af615e7347b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-26 23:53:01.000000000","message":"Patch Set 44:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/44/check/check-ibm-tempest-dsvm-full/bb3d4d5/ : SUCCESS in 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"059d418652356016b583164a63f6d1a8a00fba3b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-26 23:56:05.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/44/ : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":44},{"id":"332d2636a9285d6cd97cf44d52c2f6a468d21d1c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-26 23:59:31.000000000","message":"Patch Set 44:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/44/check/check-dsvm-tempest-vz7-exe-minimal/467603b : FAILURE in 53m 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":44},{"id":"2aac1293bb7cd2d058a0ea4e17597db99d6d9c4a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-27 00:03:12.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/44/check/gate-nova-pep8/d7c1be0/ : SUCCESS in 9m 02s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/44/check/gate-nova-docs/14b3b18//doc/build/html/ : SUCCESS in 14m 03s\n- gate-nova-python27 http://logs.openstack.org/35/193135/44/check/gate-nova-python27/3ef056f/ : SUCCESS in 15m 27s\n- gate-nova-python34 http://logs.openstack.org/35/193135/44/check/gate-nova-python34/2b8d2ee/ : SUCCESS in 12m 52s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-full/cbbeb1e/ : SUCCESS in 43m 11s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-postgres-full/7e41ffa/ : SUCCESS in 52m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-neutron-full/c28bd20/ : SUCCESS in 54m 50s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/44/check/gate-grenade-dsvm/096b75e/ : SUCCESS in 40m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-large-ops/39abd5d/ : SUCCESS in 27m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-neutron-large-ops/07e13f2/ : SUCCESS in 26m 47s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/44/check/gate-nova-tox-functional/d89d32a/ : SUCCESS in 16m 06s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/44/check/gate-grenade-dsvm-multinode/e3698fc/ : SUCCESS in 45m 19s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3f03ddd/ : SUCCESS in 35m 07s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-nova-v20-api/4bccfcb/ : SUCCESS in 43m 39s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-nova-v20-api-legacy/37193a4/ : SUCCESS in 35m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-cells/5a77c52/ : SUCCESS in 39m 39s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-full-ceph/afda44f/ : SUCCESS in 50m 47s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/7166b31/ : SUCCESS in 45m 12s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/44/check/gate-tempest-dsvm-multinode-full/b93fa6f/ : FAILURE in 52m 51s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/44/check/gate-compute-api-guide/6845ec7//api-guide/build/html/ : SUCCESS in 8m 35s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/44/check/gate-nova-commit-message/fb8d994/ : SUCCESS in 1m 20s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/44/check/gate-nova-python27-constraints/38b368a/ : SUCCESS in 15m 59s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/44/check/gate-nova-python34-constraints/41620d5/ : SUCCESS in 14m 06s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/44/check/gate-nova-docs-constraints/2c7c98e//doc/build/html/ : SUCCESS in 14m 05s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/44/check/gate-nova-pep8-constraints/a2cf235/ : SUCCESS in 10m 05s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/44/check/gate-nova-tox-constraints-functional-constraints/06daf18/ : SUCCESS in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"829ffd398aa99c7514dbdad9782a4bafb82757c2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-27 00:06:16.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/44/Nova-ML2-Sriov : SUCCESS in 1h 00m 02s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":44},{"id":"82e6a09c3039d057cea929ffeac4c8319443261f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-27 00:14:37.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/44/check/dsvm-tempest-xen/558b7f6 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":44},{"id":"e34fc2c979022b877baa914532c8b7c87a39bcec","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-27 14:30:31.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.","accounts_in_message":[],"_revision_number":45},{"id":"2b03a6eec608baf5ed8965b406ff0c1053e91441","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-27 14:48:50.000000000","message":"Patch Set 45:\n\n* pci-test http://52.27.155.124/pci/193135/45 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"461fc7ef6df5f956c96941df8144c8f0b02b18da","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-27 14:55:16.000000000","message":"Patch Set 45: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/45/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0dbedd4b156a43548af278c3d29114d3/index.html : SUCCESS in 12m 40s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/45/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a6703e13c5fe41d5b0638ae2d4b6c97b/index.html : SUCCESS in 15m 53s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/45/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e07cd031629c4257bddfe13d611c253a/index.html : SUCCESS in 14m 46s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/45/check/real-db-upgrade_nova_percona_user_001:th-percona/7a009975270e4c7a8398bef0c9847d87/index.html : SUCCESS in 16m 14s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/45/check/real-db-upgrade_nova_mysql_user_002:th-mysql/29ec57327bb941d8bb3576b100814cf5/index.html : SUCCESS in 12m 39s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/45/check/real-db-upgrade_nova_percona_user_002:th-percona/e7c1d7c766d74f26a82da2913a156c08/index.html : SUCCESS in 14m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":45},{"id":"828f003d16cd1f28840c03ae4448d5db553f9dff","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-27 15:00:30.000000000","message":"Patch Set 45:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/45/ : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":45},{"id":"cf971b3cf0de444954d1e196afa756b2e6955ca6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-27 15:21:45.000000000","message":"Patch Set 45:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/45/check/check-ibm-tempest-dsvm-full/9502420/ : SUCCESS in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"eae3f4ebda8e0e4e4e9d25c4838914585136db3d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-27 15:23:03.000000000","message":"Patch Set 45:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/45/check/check-dsvm-tempest-vz7-exe-minimal/805db3c : FAILURE in 49m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":45},{"id":"52938f1a0ff510fb20e4e940edd362273cdb4974","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-27 15:31:11.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/45/check/gate-nova-pep8/bb3592a/ : SUCCESS in 10m 50s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/45/check/gate-nova-docs/3f0294e//doc/build/html/ : SUCCESS in 15m 52s\n- gate-nova-python27 http://logs.openstack.org/35/193135/45/check/gate-nova-python27/4022b89/ : SUCCESS in 14m 46s\n- gate-nova-python34 http://logs.openstack.org/35/193135/45/check/gate-nova-python34/6ede987/ : SUCCESS in 12m 52s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-full/25a8abd/ : SUCCESS in 43m 18s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-postgres-full/b6d6102/ : SUCCESS in 47m 47s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-neutron-full/a0656ac/ : SUCCESS in 58m 23s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/45/check/gate-grenade-dsvm/e216ae2/ : SUCCESS in 37m 46s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-large-ops/42dde69/ : SUCCESS in 39m 03s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-neutron-large-ops/e389d4a/ : SUCCESS in 27m 28s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/45/check/gate-nova-tox-functional/a84ac3c/ : SUCCESS in 16m 18s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/45/check/gate-grenade-dsvm-multinode/d554f34/ : SUCCESS in 54m 29s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/286462c/ : SUCCESS in 40m 16s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-nova-v20-api/2664fa7/ : SUCCESS in 35m 32s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-nova-v20-api-legacy/9b9c5e3/ : SUCCESS in 36m 42s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-cells/befaed7/ : SUCCESS in 48m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-full-ceph/02d62b2/ : SUCCESS in 53m 54s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/97dab43/ : SUCCESS in 52m 14s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/45/check/gate-tempest-dsvm-multinode-full/ec49c4a/ : FAILURE in 54m 24s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/45/check/gate-compute-api-guide/25e683d//api-guide/build/html/ : SUCCESS in 9m 18s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/45/check/gate-nova-commit-message/3a94503/ : SUCCESS in 1m 23s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/45/check/gate-nova-python27-constraints/dd19772/ : SUCCESS in 14m 51s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/45/check/gate-nova-python34-constraints/0073574/ : SUCCESS in 13m 31s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/45/check/gate-nova-docs-constraints/87f599c//doc/build/html/ : SUCCESS in 14m 42s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/45/check/gate-nova-pep8-constraints/c3f1588/ : SUCCESS in 7m 24s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/45/check/gate-nova-tox-constraints-functional-constraints/706133f/ : SUCCESS in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"de03da93cf8fab979b763e135cf53a06250c848a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-27 15:32:23.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/45/Nova-ML2-Sriov : SUCCESS in 58m 19s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":45},{"id":"a05f2478d3cc2c0070acaeb7b1ed08a40cbc1bed","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-27 15:34:14.000000000","message":"Patch Set 45: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/45/21068/index.html : FAILURE in 55 m 26 s","accounts_in_message":[],"_revision_number":45},{"id":"71d64b65d9444a1a221bb0face34d0002ded4cf2","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-27 16:31:57.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/45/check/dsvm-tempest-xen/7ba0448 : FAILURE in 1h 59m 06s","accounts_in_message":[],"_revision_number":45},{"id":"ba05a1d70de88cf41d8af0c6775d46d60578671e","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-27 19:48:42.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.","accounts_in_message":[],"_revision_number":46},{"id":"9eef7b5029c78f95f621f6027348fc4afeff5d3f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-27 20:06:06.000000000","message":"Patch Set 46:\n\n* pci-test http://52.27.155.124/pci/193135/46 : SUCCESS","accounts_in_message":[],"_revision_number":46},{"id":"5304cee405ee09b7ead6c38e4415d042ce7e1851","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-27 20:13:29.000000000","message":"Patch Set 46: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/46/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7b6c2c9cc233497bbc5878ce0b9061fd/index.html : SUCCESS in 12m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/46/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b8d4b942e12c42ac890a859c495bf79c/index.html : SUCCESS in 16m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/46/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/145b3ad3da09420d884e844e2538da60/index.html : SUCCESS in 12m 23s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/46/check/real-db-upgrade_nova_percona_user_001:th-percona/856e3cf03d77459397cda8ce13af1c28/index.html : SUCCESS in 16m 05s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/46/check/real-db-upgrade_nova_mysql_user_002:th-mysql/492459d1ff8f4960bc07e5b379359fd5/index.html : SUCCESS in 12m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/46/check/real-db-upgrade_nova_percona_user_002:th-percona/e83a1a13b0984148a1ceffaa7dd72a20/index.html : SUCCESS in 14m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":46},{"id":"c88e2a70e2e2b8895e8196ce9092c8ef235d7d33","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-27 20:38:37.000000000","message":"Patch Set 46:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/46/check/check-dsvm-tempest-vz7-exe-minimal/1bbacb0 : SUCCESS in 46m 47s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":46},{"id":"013d9faefd458beab0ba275a6e743c40cd18f3c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-27 20:40:05.000000000","message":"Patch Set 46:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/46/check/check-ibm-tempest-dsvm-full/2d25012/ : FAILURE in 48m 55s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"234f41ce584e18d64067c82c9c744746345183c5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-27 20:50:22.000000000","message":"Patch Set 46:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/46/Nova-ML2-Sriov : SUCCESS in 58m 28s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":46},{"id":"137fe429bddae72b08a281f63f010acc20c63f5b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-27 20:57:08.000000000","message":"Patch Set 46: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/46/21077/index.html : FAILURE in 01 h 05 m 51 s","accounts_in_message":[],"_revision_number":46},{"id":"9adfd145572b7b1f0bc1209597ad9949478db59c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-27 21:00:00.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/46/check/gate-nova-pep8/9c44eb6/ : FAILURE in 9m 39s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/46/check/gate-nova-docs/9685624//doc/build/html/ : SUCCESS in 15m 02s\n- gate-nova-python27 http://logs.openstack.org/35/193135/46/check/gate-nova-python27/72dcd84/ : FAILURE in 14m 20s\n- gate-nova-python34 http://logs.openstack.org/35/193135/46/check/gate-nova-python34/338fe6c/ : SUCCESS in 14m 15s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-full/e70227a/ : FAILURE in 52m 30s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-postgres-full/95d70f4/ : FAILURE in 49m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-neutron-full/bf3ca24/ : FAILURE in 1h 09m 01s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/46/check/gate-grenade-dsvm/36e380e/ : SUCCESS in 38m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-large-ops/b26c761/ : SUCCESS in 28m 07s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-neutron-large-ops/d486f34/ : SUCCESS in 29m 05s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/46/check/gate-nova-tox-functional/dd111bc/ : SUCCESS in 15m 14s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/46/check/gate-grenade-dsvm-multinode/3eeafb3/ : FAILURE in 41m 56s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3b26e40/ : SUCCESS in 38m 20s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-nova-v20-api/072aadc/ : FAILURE in 40m 19s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-nova-v20-api-legacy/e5f6de7/ : FAILURE in 55m 45s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-cells/cd3e123/ : FAILURE in 1h 08m 02s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-full-ceph/725fa79/ : FAILURE in 1h 01m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4590c6b/ : FAILURE in 51m 59s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/46/check/gate-tempest-dsvm-multinode-full/f5acd9d/ : FAILURE in 1h 03m 24s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/46/check/gate-compute-api-guide/4889b2b//api-guide/build/html/ : SUCCESS in 8m 48s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/46/check/gate-nova-commit-message/494745d/ : SUCCESS in 1m 23s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/46/check/gate-nova-python27-constraints/0947045/ : FAILURE in 15m 15s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/46/check/gate-nova-python34-constraints/c96557b/ : SUCCESS in 13m 13s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/46/check/gate-nova-docs-constraints/c6f90f7//doc/build/html/ : SUCCESS in 14m 49s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/46/check/gate-nova-pep8-constraints/bf0ad02/ : FAILURE in 9m 26s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/46/check/gate-nova-tox-constraints-functional-constraints/1c0b3ee/ : FAILURE in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"237786d8974444c9afb4fe10fc9c2b7904e3866d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-27 21:00:52.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/46/check/dsvm-tempest-xen/4bbdfe3 : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":46},{"id":"b8a6ca9203fdf2ae65649b9115527a3bd3b89266","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-27 21:18:58.000000000","message":"Patch Set 46:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/46/ : FAILURE in 1h 27m 43s","accounts_in_message":[],"_revision_number":46},{"id":"0f33ac7e8a91200f747a45ce9a193fc06cec4184","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-27 22:23:53.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.","accounts_in_message":[],"_revision_number":47},{"id":"a98925b4c50cb25182f2dbfde4ccea9436de2e86","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-27 22:41:26.000000000","message":"Patch Set 47:\n\n* pci-test http://52.27.155.124/pci/193135/47 : SUCCESS","accounts_in_message":[],"_revision_number":47},{"id":"7121cd35b22164273cb0acf398a1ec9313b148ea","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-27 22:49:19.000000000","message":"Patch Set 47: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/47/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3a2a7a19633c4f3993a80efb18511df3/index.html : SUCCESS in 14m 45s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/47/check/real-db-upgrade_nova_mysql_user_001:th-mysql/23be8ef5c5c1417983be8132f6cfc4eb/index.html : SUCCESS in 16m 33s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/47/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/cdcc0582984a416998a4facc735bec30/index.html : SUCCESS in 14m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/47/check/real-db-upgrade_nova_percona_user_001:th-percona/3d820a3e204c407897d2b1e7d8a48118/index.html : SUCCESS in 16m 29s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/47/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5fa354ce08aa4facbf237db36b931179/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/47/check/real-db-upgrade_nova_percona_user_002:th-percona/a98af873799448c48bb8faed12f02233/index.html : SUCCESS in 14m 59s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":47},{"id":"ef84e50657a9332c18841169ead1c0c2cf0b8c2d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-27 23:13:59.000000000","message":"Patch Set 47:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/47/check/check-ibm-tempest-dsvm-full/6fb7fc5/ : FAILURE in 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"b93ac73190a46146d76bba472434aa581869a6aa","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-27 23:14:48.000000000","message":"Patch Set 47:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/47/check/check-dsvm-tempest-vz7-exe-minimal/291d598 : FAILURE in 48m 09s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":47},{"id":"0ab845126d476603e3718fd699e0ab1e79a8db18","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-27 23:22:59.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/47/Nova-ML2-Sriov : SUCCESS in 56m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":47},{"id":"df737c0b5e93cb10322248c169feaa0c92bfc551","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-27 23:24:10.000000000","message":"Patch Set 47: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/47/21084/index.html : FAILURE in 57 m 51 s","accounts_in_message":[],"_revision_number":47},{"id":"92941f6d9a15f9dd98ad3742818ffd4f8ca81fda","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-27 23:35:30.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/47/check/gate-nova-pep8/0a98fd2/ : SUCCESS in 9m 29s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/47/check/gate-nova-docs/5790779//doc/build/html/ : SUCCESS in 14m 15s\n- gate-nova-python27 http://logs.openstack.org/35/193135/47/check/gate-nova-python27/1ef3f41/ : SUCCESS in 14m 37s\n- gate-nova-python34 http://logs.openstack.org/35/193135/47/check/gate-nova-python34/e2a9402/ : SUCCESS in 13m 09s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-full/05ab6f8/ : FAILURE in 51m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-postgres-full/1aafdde/ : FAILURE in 53m 01s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-neutron-full/43a7b21/ : FAILURE in 1h 09m 07s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/47/check/gate-grenade-dsvm/ffa6960/ : SUCCESS in 39m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-large-ops/bfe6443/ : SUCCESS in 29m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-neutron-large-ops/d464afa/ : SUCCESS in 29m 13s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/47/check/gate-nova-tox-functional/be689dc/ : SUCCESS in 15m 47s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/47/check/gate-grenade-dsvm-multinode/6c59cc1/ : SUCCESS in 54m 36s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f295a51/ : SUCCESS in 38m 20s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-nova-v20-api/0e677a2/ : FAILURE in 40m 19s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-nova-v20-api-legacy/70a1622/ : FAILURE in 41m 36s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-cells/1e0725d/ : FAILURE in 56m 58s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-full-ceph/7a325e3/ : FAILURE in 1h 02m 03s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/2e9abee/ : FAILURE in 1h 06m 25s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/47/check/gate-tempest-dsvm-multinode-full/88326ab/ : FAILURE in 1h 04m 12s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/47/check/gate-compute-api-guide/7f294d2//api-guide/build/html/ : SUCCESS in 8m 40s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/47/check/gate-nova-commit-message/e99ed4a/ : SUCCESS in 1m 21s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/47/check/gate-nova-python27-constraints/ae11c9a/ : SUCCESS in 14m 42s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/47/check/gate-nova-python34-constraints/367ca2c/ : SUCCESS in 10m 15s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/47/check/gate-nova-docs-constraints/e7931f5//doc/build/html/ : SUCCESS in 12m 00s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/47/check/gate-nova-pep8-constraints/89cce3c/ : SUCCESS in 9m 08s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/47/check/gate-nova-tox-constraints-functional-constraints/b32b7ec/ : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"60cb989f52b5b9dd522ed280580f7c9151fc07d1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-27 23:36:05.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/47/check/dsvm-tempest-xen/5421e38 : SUCCESS in 1h 10m 04s","accounts_in_message":[],"_revision_number":47},{"id":"e5de315821d9b1cb8e1082a7400e05fbb54fc8c8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-27 23:56:33.000000000","message":"Patch Set 47:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/47/ : FAILURE in 1h 30m 11s","accounts_in_message":[],"_revision_number":47},{"id":"7e5e6d4c104a9111535bbef2452a297cd0272ae8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-28 10:38:39.000000000","message":"Patch Set 38: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/38 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/612/ : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"669d66e3a53aa0045a1a3ca38ee92730cd1387c7","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-28 15:43:22.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.","accounts_in_message":[],"_revision_number":48},{"id":"01925960d0512b441e3a3c09fe3a87284a09ab17","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-28 16:05:01.000000000","message":"Patch Set 48:\n\n* pci-test http://52.27.155.124/pci/193135/48 : SUCCESS","accounts_in_message":[],"_revision_number":48},{"id":"c248648a28960cb3dff776e2a6d343e16b7273fe","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-28 16:41:40.000000000","message":"Patch Set 48: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/48/21143/index.html : SUCCESS in 55 m 23 s","accounts_in_message":[],"_revision_number":48},{"id":"c2bfaee10b0a412aa014802fed64a994baf07156","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-28 16:44:19.000000000","message":"Patch Set 48: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/48/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/cd7206d443be49ed8fe600e448a9d467/index.html : SUCCESS in 13m 07s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/48/check/real-db-upgrade_nova_mysql_user_001:th-mysql/64e3ec42f464403f8e64ad4e42e8e6d1/index.html : SUCCESS in 16m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/48/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dfea1a2dedc3442b941cde8fb276dbd0/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/48/check/real-db-upgrade_nova_percona_user_001:th-percona/3193fe736e7c40bfabfa4d161253f14c/index.html : SUCCESS in 16m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/48/check/real-db-upgrade_nova_mysql_user_002:th-mysql/56365640ff2e4e368bb320c37e2cbecc/index.html : SUCCESS in 14m 22s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/48/check/real-db-upgrade_nova_percona_user_002:th-percona/f0422e480f824abc872ffe2c33cf8ee5/index.html : SUCCESS in 13m 11s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":48},{"id":"b4b92f124bf2393e97ecf3dde2ce9967b6253cf9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-28 16:53:11.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/48/check/gate-nova-pep8/b3313b3/ : SUCCESS in 10m 02s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/48/check/gate-nova-docs/f018738//doc/build/html/ : SUCCESS in 14m 11s\n- gate-nova-python27 http://logs.openstack.org/35/193135/48/check/gate-nova-python27/65cfd09/ : SUCCESS in 14m 40s\n- gate-nova-python34 http://logs.openstack.org/35/193135/48/check/gate-nova-python34/0e6ec38/ : SUCCESS in 17m 42s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-full/7130732/ : FAILURE in 45m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-postgres-full/d83adc1/ : FAILURE in 52m 38s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-neutron-full/4111710/ : FAILURE in 1h 02m 08s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/48/check/gate-grenade-dsvm/d2923f7/ : SUCCESS in 41m 22s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-large-ops/73729c0/ : SUCCESS in 29m 03s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-neutron-large-ops/2426dc7/ : SUCCESS in 27m 19s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/48/check/gate-nova-tox-functional/ab8cece/ : SUCCESS in 15m 37s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/48/check/gate-grenade-dsvm-multinode/af3892f/ : SUCCESS in 47m 23s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3a8efa4/ : SUCCESS in 39m 17s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-nova-v20-api/e393278/ : SUCCESS in 46m 48s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-nova-v20-api-legacy/d4f0220/ : SUCCESS in 41m 55s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-cells/e9da14d/ : SUCCESS in 45m 44s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-full-ceph/3cd5ff8/ : FAILURE in 55m 39s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5293892/ : FAILURE in 52m 41s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/48/check/gate-tempest-dsvm-multinode-full/206a7cd/ : FAILURE in 52m 40s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/48/check/gate-compute-api-guide/5594876//api-guide/build/html/ : SUCCESS in 8m 45s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/48/check/gate-nova-commit-message/fd307ce/ : SUCCESS in 1m 16s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/48/check/gate-nova-python27-constraints/41793a3/ : SUCCESS in 12m 26s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/48/check/gate-nova-python34-constraints/52b1ee2/ : SUCCESS in 11m 05s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/48/check/gate-nova-docs-constraints/5519123//doc/build/html/ : SUCCESS in 12m 11s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/48/check/gate-nova-pep8-constraints/502828b/ : SUCCESS in 9m 02s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/48/check/gate-nova-tox-constraints-functional-constraints/90460e6/ : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"4a84f2ee4f9408ce06f55003890081743d9a5ab9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-28 16:58:16.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/48/ : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":48},{"id":"6fdab0ca8d650a40f4b18c86d3bc2a80db640c55","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-28 17:06:20.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/48/check/dsvm-tempest-xen/d0c71e6 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":48},{"id":"84e31f3ac68e8ec510fa0c06dbf0f3a73b630527","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-28 17:08:39.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"63d294f78ed0144906baaacb9fed66b9b7650998","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-28 17:28:39.000000000","message":"Patch Set 49:\n\n* pci-test http://52.27.155.124/pci/193135/49 : SUCCESS","accounts_in_message":[],"_revision_number":49},{"id":"ec75026b14dce2f575cbaa0fa52de8c5f9d36243","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-28 18:04:20.000000000","message":"Patch Set 49:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/49/check/check-ibm-tempest-dsvm-full/255f79c/ : SUCCESS in 49m 12s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"b9f71b2813cc985f4f5ed5df14f70069915b6626","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-28 18:06:27.000000000","message":"Patch Set 49: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/49/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/56a2c8f7fcb040559e579b2cd10cccc2/index.html : SUCCESS in 12m 12s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/49/check/real-db-upgrade_nova_mysql_user_001:th-mysql/49fe914999274df4946fbac4892c76c5/index.html : SUCCESS in 15m 46s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/49/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/12500eb3d31a4008945d02ec2e2765c6/index.html : SUCCESS in 13m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/49/check/real-db-upgrade_nova_percona_user_001:th-percona/f93456b003774b6ab9162a62f2369515/index.html : SUCCESS in 14m 12s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/49/check/real-db-upgrade_nova_mysql_user_002:th-mysql/639c75bcdf4948d79f5652ba7f9a2f94/index.html : SUCCESS in 12m 22s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/49/check/real-db-upgrade_nova_percona_user_002:th-percona/3a0cf5948c5446ce935aab5ba9d9a609/index.html : SUCCESS in 15m 02s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":49},{"id":"14327edea07e4cf3cc5e2eb1832b43f26c4992c3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-28 18:11:11.000000000","message":"Patch Set 49: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/49/21150/index.html : SUCCESS in 58 m 48 s","accounts_in_message":[],"_revision_number":49},{"id":"5fad570e7dc8432eb93de096b2ce4cbcc238fe69","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-28 18:23:43.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/49/ : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":49},{"id":"a7de2558101e6ca2ffb23c42f101887e0df3fe87","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-28 18:28:33.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/49/check/dsvm-tempest-xen/5731e0d : FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":49},{"id":"98392e6b9e25d2149db6d47d010e2edc88edad83","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-28 18:32:30.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/49/check/gate-nova-pep8/07a5121/ : SUCCESS in 6m 58s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/49/check/gate-nova-docs/edace2f//doc/build/html/ : SUCCESS in 15m 29s\n- gate-nova-python27 http://logs.openstack.org/35/193135/49/check/gate-nova-python27/0b4fa78/ : SUCCESS in 15m 47s\n- gate-nova-python34 http://logs.openstack.org/35/193135/49/check/gate-nova-python34/3c1713f/ : SUCCESS in 12m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-full/bdc0bb2/ : SUCCESS in 1h 02m 32s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-postgres-full/f90a57c/ : SUCCESS in 56m 00s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-neutron-full/15234f4/ : SUCCESS in 1h 09m 04s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/49/check/gate-grenade-dsvm/bbf9fed/ : SUCCESS in 54m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-large-ops/9a2152a/ : SUCCESS in 36m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-neutron-large-ops/6959334/ : SUCCESS in 31m 42s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/49/check/gate-nova-tox-functional/8e79d0f/ : SUCCESS in 12m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/49/check/gate-grenade-dsvm-multinode/0d7075e/ : SUCCESS in 51m 33s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/571efdf/ : SUCCESS in 49m 11s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-nova-v20-api/5710f70/ : SUCCESS in 51m 43s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-nova-v20-api-legacy/9e3a804/ : SUCCESS in 47m 11s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-cells/f6009cc/ : SUCCESS in 36m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-full-ceph/60f6235/ : SUCCESS in 45m 03s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/db936e2/ : SUCCESS in 47m 22s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/49/check/gate-tempest-dsvm-multinode-full/6a576eb/ : FAILURE in 49m 20s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/49/check/gate-compute-api-guide/5ca6ec1//api-guide/build/html/ : SUCCESS in 8m 40s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/49/check/gate-nova-commit-message/cd6d988/ : SUCCESS in 1m 26s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/49/check/gate-nova-python27-constraints/f8f03af/ : SUCCESS in 14m 54s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/49/check/gate-nova-python34-constraints/63ab5c0/ : SUCCESS in 13m 15s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/49/check/gate-nova-docs-constraints/4be2d09//doc/build/html/ : SUCCESS in 14m 29s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/49/check/gate-nova-pep8-constraints/7c45a9a/ : SUCCESS in 11m 59s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/49/check/gate-nova-tox-constraints-functional-constraints/d80de47/ : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"8aed7cd78b05df169edd1aaf54b6e409fe2d1fe5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-28 19:05:24.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/49/check/check-dsvm-tempest-vz7-exe-minimal/d3dcf25 : SUCCESS in 44m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":49},{"id":"55c47fe60addb95854b1266f91604747a3277989","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-28 19:06:08.000000000","message":"Patch Set 49:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/49/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":49},{"id":"88d24ac5e28dca6f838afdb265ea3a47d02c982b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-28 19:08:43.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/49/Nova-ML2-Sriov : SUCCESS in 54m 30s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":49},{"id":"ec6213b6ad875bd765ba95ed706d437ed7cda19c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-29 12:09:28.000000000","message":"Patch Set 49:\n\nrecheck","accounts_in_message":[],"_revision_number":49},{"id":"c335b262db34eb0e82f75a7361bb7bf5a6ad8ffa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-29 13:17:52.000000000","message":"Patch Set 49:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/49/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":49},{"id":"35ac1a1285faac364d22f6d5d48c49c346ff53ee","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-29 13:44:09.000000000","message":"Patch Set 49:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/49/ : FAILURE in 1h 21m 37s","accounts_in_message":[],"_revision_number":49},{"id":"d728ac39357e7ac692c00969945f1868a58ff1c2","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-29 13:55:27.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.","accounts_in_message":[],"_revision_number":50},{"id":"d73540547604ec395df8d86575a862308e232df2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-29 14:18:43.000000000","message":"Patch Set 50:\n\n* pci-test http://52.27.155.124/pci/193135/50 : SUCCESS","accounts_in_message":[],"_revision_number":50},{"id":"3da4bfcf953056a99ad4bc82d810e5f70f3d0704","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-29 14:19:20.000000000","message":"Patch Set 50:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":50},{"id":"f140580263054383f2da4dd91cd79d23fe719458","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-29 14:19:24.000000000","message":"Patch Set 50: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":50},{"id":"84e23b3d2ab300a6409cb304ab78f74a7c2b0aab","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-29 14:19:54.000000000","message":"Patch Set 50:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":50},{"id":"ed34fc3b4360e58d3c99ded93ce117a6ef505d04","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-29 14:46:45.000000000","message":"Patch Set 50: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/50/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b5f65af24d6f4fd0b6ea659ba6c1adf2/index.html : SUCCESS in 16m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/50/check/real-db-upgrade_nova_mysql_user_001:th-mysql/04c58b5ce9c94fe0ab0e4c5a3f43c2b4/index.html : SUCCESS in 13m 55s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/50/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c1d04d8f7a004244afff0651b1a1b36a/index.html : SUCCESS in 15m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/50/check/real-db-upgrade_nova_percona_user_001:th-percona/7b8ea0a8d82b4337bb2b9b6b1a579861/index.html : SUCCESS in 15m 53s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/50/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5e093605ff14427aa3d50236c3aabbdf/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/50/check/real-db-upgrade_nova_percona_user_002:th-percona/af3fa4589f944a84806e191878d3cd75/index.html : SUCCESS in 14m 58s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":50},{"id":"265f26f01e048145228acb03b5710a6158444b8c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-29 15:13:16.000000000","message":"Patch Set 50: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/50/21193/index.html : FAILURE in 01 h 14 m 04 s","accounts_in_message":[],"_revision_number":50},{"id":"b5a77bc402d7cf27af93f78d471248bb1b61dcf8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-29 15:29:05.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/50/check/gate-nova-pep8/17f4dee/ : SUCCESS in 10m 05s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/50/check/gate-nova-docs/419b71f//doc/build/html/ : SUCCESS in 14m 24s\n- gate-nova-python27 http://logs.openstack.org/35/193135/50/check/gate-nova-python27/e93ffe5/ : SUCCESS in 14m 13s\n- gate-nova-python34 http://logs.openstack.org/35/193135/50/check/gate-nova-python34/3bb9e44/ : FAILURE in 29m 36s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-full/d6ef207/ : SUCCESS in 52m 34s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-postgres-full/9cc0c5b/ : SUCCESS in 39m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-neutron-full/c5e060b/ : SUCCESS in 52m 57s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/50/check/gate-grenade-dsvm/9aa8a42/ : SUCCESS in 42m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-large-ops/ac12f91/ : SUCCESS in 35m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-neutron-large-ops/c964686/ : SUCCESS in 38m 30s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/50/check/gate-nova-tox-functional/b2385dd/ : SUCCESS in 15m 59s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/50/check/gate-grenade-dsvm-multinode/a2a4064/ : SUCCESS in 46m 50s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/5663c0f/ : SUCCESS in 43m 32s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-nova-v20-api/864befb/ : SUCCESS in 45m 57s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-nova-v20-api-legacy/56a2df5/ : SUCCESS in 42m 49s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-cells/daaad9c/ : SUCCESS in 47m 29s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-full-ceph/b016ead/ : SUCCESS in 51m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/6e0d64b/ : SUCCESS in 45m 03s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-multinode-full/94bc8bb/ : SUCCESS in 57m 23s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/50/check/gate-compute-api-guide/5e0da9a//api-guide/build/html/ : SUCCESS in 9m 03s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/50/check/gate-nova-commit-message/77515ff/ : SUCCESS in 1m 21s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-python27-constraints/db74478/ : SUCCESS in 17m 11s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-python34-constraints/38975c5/ : SUCCESS in 31m 10s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/50/check/gate-nova-docs-constraints/d482f3b//doc/build/html/ : SUCCESS in 14m 53s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-pep8-constraints/7f284f4/ : SUCCESS in 10m 10s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-tox-constraints-functional-constraints/3398704/ : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"15e2cbf873f8ebc0630ce3e3999694fe1aa93c3e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-29 15:35:31.000000000","message":"Patch Set 50:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/50/Nova-ML2-Sriov : SUCCESS in 28m 25s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":50},{"id":"d9a3606f05d4985e6f3e28e5d46f39b1434989d3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-29 16:09:52.000000000","message":"Patch Set 50:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/50/check/check-dsvm-tempest-vz7-exe-minimal/2e058db : FAILURE in 20s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":50},{"id":"609b7fae1dc921f015c73287cff2464874a5ae28","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-29 16:28:33.000000000","message":"Patch Set 50:\n\nrecheck","accounts_in_message":[],"_revision_number":50},{"id":"548f7c008f3d64f52f599ecac937092bd483643e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-29 17:22:50.000000000","message":"Patch Set 50:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/50/check/check-ibm-tempest-dsvm-full/2d559e3/ : FAILURE in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"936596998a8050a57ff8695cc4819e9351c794ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-29 18:28:56.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/50/check/gate-nova-pep8/e0b0c66/ : SUCCESS in 6m 53s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/50/check/gate-nova-docs/982d526//doc/build/html/ : SUCCESS in 12m 11s\n- gate-nova-python27 http://logs.openstack.org/35/193135/50/check/gate-nova-python27/af8b9f9/ : SUCCESS in 11m 15s\n- gate-nova-python34 http://logs.openstack.org/35/193135/50/check/gate-nova-python34/e4cf284/ : SUCCESS in 25m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-full/a7ffecf/ : SUCCESS in 51m 41s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-postgres-full/6172f60/ : SUCCESS in 49m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-neutron-full/b17051e/ : SUCCESS in 57m 53s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/50/check/gate-grenade-dsvm/93f209f/ : SUCCESS in 53m 02s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-large-ops/e1b2ff2/ : SUCCESS in 33m 25s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-neutron-large-ops/549d12b/ : SUCCESS in 32m 30s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/50/check/gate-nova-tox-functional/9182aa6/ : SUCCESS in 17m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/50/check/gate-grenade-dsvm-multinode/4c160a3/ : SUCCESS in 57m 54s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c549471/ : SUCCESS in 45m 40s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-nova-v20-api/09cbaa7/ : SUCCESS in 40m 02s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-nova-v20-api-legacy/59101ab/ : SUCCESS in 41m 02s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-cells/74539cf/ : SUCCESS in 48m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-full-ceph/6e45711/ : SUCCESS in 52m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4930872/ : SUCCESS in 43m 05s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/50/check/gate-tempest-dsvm-multinode-full/4ab605b/ : SUCCESS in 54m 36s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/50/check/gate-compute-api-guide/05e67ac//api-guide/build/html/ : SUCCESS in 7m 01s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/50/check/gate-nova-commit-message/3319be3/ : SUCCESS in 1m 34s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-python27-constraints/1ffde37/ : SUCCESS in 14m 28s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-python34-constraints/687c62a/ : SUCCESS in 27m 10s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/50/check/gate-nova-docs-constraints/8cb0996//doc/build/html/ : SUCCESS in 16m 29s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-pep8-constraints/3fdbeeb/ : SUCCESS in 8m 04s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/50/check/gate-nova-tox-constraints-functional-constraints/17b657d/ : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"b7702ee1f756ce7433dd9fc851457feeebdcc29f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-29 18:44:33.000000000","message":"Patch Set 50:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/50/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":50},{"id":"b9b154a477c52283c1f69618d0658260d7133fd6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-29 19:49:46.000000000","message":"Patch Set 50:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/50/ : FAILURE in 3h 20m 14s","accounts_in_message":[],"_revision_number":50},{"id":"10b00e4b387775f4995e75c872508058027313a8","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2015-12-30 07:59:27.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.","accounts_in_message":[],"_revision_number":51},{"id":"7bf38908a5923d0769ec830fd9c659cb9f963915","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-30 08:01:56.000000000","message":"Patch Set 51:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/51/check/check-dsvm-tempest-vz7-exe-minimal/04a61e7 : FAILURE in 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":51},{"id":"b997affe882bbba5ed88794caf13b17fe0a15090","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-30 08:16:58.000000000","message":"Patch Set 51:\n\n* pci-test http://52.27.155.124/pci/193135/51 : SUCCESS","accounts_in_message":[],"_revision_number":51},{"id":"c4e3e9722a14e3d474ccf6b9437c8a34f556d317","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-30 08:21:04.000000000","message":"Patch Set 51: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/51/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4714128f01e04dc4b7a63f2956e93f65/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/51/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8abe98dfa01e4a17b6bc0896a9ad2809/index.html : SUCCESS in 13m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/51/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/bbec1b658621476487d07dd4ca05fa52/index.html : SUCCESS in 14m 45s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/51/check/real-db-upgrade_nova_percona_user_001:th-percona/cc9d8d13f3934d07a913dea2218b091d/index.html : SUCCESS in 19m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/51/check/real-db-upgrade_nova_mysql_user_002:th-mysql/9e95da918cd14e94b0e788f18ab6792c/index.html : SUCCESS in 12m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/51/check/real-db-upgrade_nova_percona_user_002:th-percona/b08fc82bee2d439faeae6f652ba097d1/index.html : SUCCESS in 14m 29s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":51},{"id":"38e5e0fa615b7d5427a8b0c7d522b28df167868f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-30 08:30:10.000000000","message":"Patch Set 51:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/51/Nova-ML2-Sriov : SUCCESS in 28m 35s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":51},{"id":"de62d68f82abb64905883b189c6e96f50e338c80","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-30 08:48:55.000000000","message":"Patch Set 51:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/51/check/check-ibm-tempest-dsvm-full/64d0133/ : SUCCESS in 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"aeaeb17e10ed348964c5f3cbf10de160fbc6616e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-30 09:04:01.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/51/check/dsvm-tempest-xen/11431de : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":51},{"id":"a19d9225dfa2c778ad882ba64ce0420f9bb2932b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-30 09:06:14.000000000","message":"Patch Set 51: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/51/21247/index.html : SUCCESS in 57 m 48 s","accounts_in_message":[],"_revision_number":51},{"id":"b6e8bf4f95c7f0585dd1d9adc5279fa2edd75e38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-30 09:12:32.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/51/check/gate-nova-pep8/de52250/ : SUCCESS in 9m 03s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/51/check/gate-nova-docs/b1dbda7//doc/build/html/ : SUCCESS in 11m 09s\n- gate-nova-python27 http://logs.openstack.org/35/193135/51/check/gate-nova-python27/618820b/ : SUCCESS in 15m 35s\n- gate-nova-python34 http://logs.openstack.org/35/193135/51/check/gate-nova-python34/be1ea85/ : SUCCESS in 29m 13s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-full/e76394b/ : SUCCESS in 51m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-postgres-full/f024508/ : SUCCESS in 53m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-neutron-full/65dc0dc/ : SUCCESS in 1h 05m 23s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/51/check/gate-grenade-dsvm/0a489aa/ : SUCCESS in 39m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-large-ops/2c96886/ : SUCCESS in 39m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-neutron-large-ops/e452700/ : SUCCESS in 36m 12s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/51/check/gate-nova-tox-functional/102cc52/ : SUCCESS in 16m 30s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/51/check/gate-grenade-dsvm-multinode/44625aa/ : SUCCESS in 48m 03s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b3adc1b/ : SUCCESS in 40m 39s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-nova-v20-api/d45ae9d/ : SUCCESS in 43m 50s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-nova-v20-api-legacy/a7e219a/ : SUCCESS in 42m 31s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-cells/579c855/ : SUCCESS in 45m 06s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-full-ceph/ffd886b/ : SUCCESS in 57m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/21c2069/ : SUCCESS in 52m 12s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/51/check/gate-tempest-dsvm-multinode-full/801af9a/ : SUCCESS in 1h 06m 57s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/51/check/gate-compute-api-guide/3b9c8f8//api-guide/build/html/ : SUCCESS in 8m 34s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/51/check/gate-nova-commit-message/dbcc066/ : SUCCESS in 1m 29s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/51/check/gate-nova-python27-constraints/08e2ef3/ : SUCCESS in 14m 52s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/51/check/gate-nova-python34-constraints/737b8e1/ : SUCCESS in 28m 51s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/51/check/gate-nova-docs-constraints/be271d7//doc/build/html/ : SUCCESS in 15m 57s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/51/check/gate-nova-pep8-constraints/c9a124c/ : SUCCESS in 7m 30s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/51/check/gate-nova-tox-constraints-functional-constraints/7356d9f/ : SUCCESS in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"f303080edcb20ad3328a910e5569c8ee49b18a0d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-30 09:51:52.000000000","message":"Patch Set 51:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/51/ : FAILURE in 1h 50m 28s","accounts_in_message":[],"_revision_number":51},{"id":"ce53541ebf836aa75d0bf43eb8af4d59019d3e70","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-03 17:57:32.000000000","message":"Patch Set 39: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/39 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/721/ : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"35bc0f77e94be812f2c7c7bbe5a7065d2a12de6f","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 07:49:54.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.","accounts_in_message":[],"_revision_number":52},{"id":"6fc57532f4a56bda2b2381e8865762eaab3c1e8a","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 08:07:37.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"422c1238b085c2e3f669d67abbd70f4c21bc58a1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-04 08:08:05.000000000","message":"Patch Set 52:\n\n* pci-test http://52.27.155.124/pci/193135/52 : SUCCESS","accounts_in_message":[],"_revision_number":52},{"id":"755bfb737faa83ca21222df8d257df624a137d85","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-04 08:31:15.000000000","message":"Patch Set 53: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/53/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/332feca0d906446382a30459a55a9289/index.html : SUCCESS in 15m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/53/check/real-db-upgrade_nova_mysql_user_001:th-mysql/fa1be86eaec54d64aaf49f0a8131d6ea/index.html : SUCCESS in 16m 20s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/53/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b1ad18176fa84897bc69dffbb52f4836/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/53/check/real-db-upgrade_nova_percona_user_001:th-percona/25ef807b984342629d41c8dbd3462c40/index.html : SUCCESS in 13m 38s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/53/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d5c8eda61c4e47109e3c071432ec0998/index.html : SUCCESS in 12m 10s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/53/check/real-db-upgrade_nova_percona_user_002:th-percona/b93002b7f21c4824b63570f92eb99275/index.html : SUCCESS in 14m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":53},{"id":"ec405b90388365a7aa760fe097a23fce2bdc3f01","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-04 08:47:42.000000000","message":"Patch Set 53:\n\n* pci-test http://52.27.155.124/pci/193135/53 : SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"3a174f07fd06418fc6fa81972ee274d2e518fd7f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-04 08:57:06.000000000","message":"Patch Set 53:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/53/check/check-ibm-tempest-dsvm-full/a351fb2/ : SUCCESS in 48m 08s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"11081e0b3061fd5c4bc08abcf2d780cf5217df75","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-04 09:07:25.000000000","message":"Patch Set 53: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/53/21450/index.html : SUCCESS in 53 m 39 s","accounts_in_message":[],"_revision_number":53},{"id":"5e6e3fa306a72795b9a99006ab34362ea59e1f69","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-04 09:24:05.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/53/check/gate-nova-pep8/f1cc779/ : SUCCESS in 10m 27s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/53/check/gate-nova-docs/6ce79ad//doc/build/html/ : SUCCESS in 13m 19s\n- gate-nova-python27 http://logs.openstack.org/35/193135/53/check/gate-nova-python27/542a975/ : SUCCESS in 15m 49s\n- gate-nova-python34 http://logs.openstack.org/35/193135/53/check/gate-nova-python34/41f9d92/ : SUCCESS in 28m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-full/679e0cb/ : SUCCESS in 56m 22s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-postgres-full/6cc5554/ : SUCCESS in 56m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-neutron-full/7497ff2/ : SUCCESS in 1h 11m 45s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/53/check/gate-grenade-dsvm/ab1e22c/ : SUCCESS in 49m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-large-ops/552d3e8/ : SUCCESS in 41m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-neutron-large-ops/06113bc/ : SUCCESS in 37m 21s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/53/check/gate-nova-releasenotes/5828369//releasenotes/build/html/ : SUCCESS in 11m 01s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/53/check/gate-nova-tox-functional/515977e/ : SUCCESS in 16m 21s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/53/check/gate-grenade-dsvm-multinode/869ea5f/ : SUCCESS in 50m 07s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/758dd18/ : SUCCESS in 42m 23s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-nova-v20-api/44d01ff/ : SUCCESS in 37m 45s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-nova-v20-api-legacy/ed11a7c/ : SUCCESS in 49m 05s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-cells/1d184a8/ : SUCCESS in 52m 54s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-full-ceph/871d53c/ : SUCCESS in 54m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5852a25/ : SUCCESS in 54m 34s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/53/check/gate-tempest-dsvm-multinode-full/902daf5/ : SUCCESS in 53m 53s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/53/check/gate-compute-api-guide/b078671//api-guide/build/html/ : SUCCESS in 9m 32s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/53/check/gate-nova-commit-message/8496f37/ : SUCCESS in 1m 39s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/53/check/gate-nova-python27-constraints/18298d7/ : SUCCESS in 14m 53s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/53/check/gate-nova-python34-constraints/adc896d/ : SUCCESS in 29m 08s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/53/check/gate-nova-docs-constraints/03186fc//doc/build/html/ : SUCCESS in 14m 28s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/53/check/gate-nova-pep8-constraints/00e53e5/ : SUCCESS in 7m 12s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/53/check/gate-nova-tox-constraints-functional-constraints/bbfca12/ : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"3bffd99785768a3b944a855fec6c6e36ecd7ab16","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-04 09:33:31.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/53/check/dsvm-tempest-xen/3170529 : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":53},{"id":"b3a831052aae561e75ac0e3f857110ad6bb4e6da","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-04 10:24:51.000000000","message":"Patch Set 53:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/53/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":53},{"id":"467edbe34cc2601a2595056624a10441b2ed58fa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-04 13:16:54.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/53/Nova-ML2-Sriov : SUCCESS in 1h 33m 14s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":53},{"id":"823fda01939c193d2e5c64f0e6c155c8dce15aca","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-01-04 13:20:54.000000000","message":"Patch Set 53: Code-Review-1\n\n(4 comments)\n\nFrom a libvirt/s390x platform view I see no issues here. I -1\u0027ed because of the testcases I would expect. As a side note: The connection info dict gets bigger and bigger and should be refactored (not in this patch) to be a real object IMO.","accounts_in_message":[],"_revision_number":53},{"id":"6ea5ed35531fe20f84773e3015b22ad18d206180","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 15:35:03.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"2de094fa469d56a8bfa4d4e794fb6836e74fa323","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-04 16:14:32.000000000","message":"Patch Set 54: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/54/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/ae9501ad003941ba86001094da37a57f/index.html : SUCCESS in 12m 49s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/54/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b0048b6945b94a0cbc90b5ce2b2c7153/index.html : SUCCESS in 16m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/54/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a4658574bff64dd58c3d677659e590ba/index.html : SUCCESS in 18m 11s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/54/check/real-db-upgrade_nova_percona_user_001:th-percona/1d6fcbb38a934f9397c0fe6ac7b62672/index.html : SUCCESS in 14m 14s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/54/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d20122c829984de08a15b3074749ea03/index.html : SUCCESS in 15m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/54/check/real-db-upgrade_nova_percona_user_002:th-percona/9331be8e039a46b8a57339f43f89bf1d/index.html : SUCCESS in 14m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":54},{"id":"dab46bc55fe344b25cf7ae3eeeeec0d6b8d3ab3b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-04 16:28:05.000000000","message":"Patch Set 54:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/54/check/check-ibm-tempest-dsvm-full/4e67b8a/ : SUCCESS in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"d38c48519a4c5fc49fb1711db0521caf2943e8af","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-04 16:41:20.000000000","message":"Patch Set 54: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/54/21546/index.html : SUCCESS in 54 m 44 s","accounts_in_message":[],"_revision_number":54},{"id":"faaa4dda7a48b21ffe9eb4f4b0282fbd15d1d0ae","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-04 16:59:43.000000000","message":"Patch Set 53: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/193135/53 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/741/ : SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"def0e85e06b0514f085ae284994ddf66a57253ce","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-04 17:15:42.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/54/check/dsvm-tempest-xen/7299b96 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":54},{"id":"bb7ebc4aefad6982b2abfec64f697928a4f81d6f","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 17:22:01.000000000","message":"Patch Set 53:\n\n(4 comments)\n\nHi Markus,\n\nThanks for the comments. Please see my comments in line. I already fixed most of them and working on the tests. I also added a question regarding the tests you mentioned, please check.\n\nThanks and Best Regards,\nIldiko","accounts_in_message":[],"_revision_number":53},{"id":"d7f96d4f7b0370b47bb67a8777a8e41eb3e13888","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 18:12:44.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.","accounts_in_message":[],"_revision_number":55},{"id":"bc4994689c2119798f600f0ef4df179d51e22dba","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 18:24:28.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.","accounts_in_message":[],"_revision_number":56},{"id":"d5c4dc512d4dcd8a7568abd7009e173b12281b79","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-04 19:13:31.000000000","message":"Patch Set 56:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/56/check/check-ibm-tempest-dsvm-full/7dcb5ca/ : SUCCESS in 47m 50s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"4a2f480ddffc5559c0afa9f22fefff03ae358bd7","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-04 19:49:19.000000000","message":"Patch Set 56: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/56/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/066353f9e294486a93f745d05762c443/index.html : SUCCESS in 12m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/56/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4cad83294a8847ddb84cdd78340912e5/index.html : SUCCESS in 17m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/56/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0e2a47f969f14f8d85da5f1da8248aa5/index.html : SUCCESS in 12m 26s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/56/check/real-db-upgrade_nova_percona_user_001:th-percona/2aefc6ebaaf04c8f8b4471935f28c712/index.html : SUCCESS in 16m 55s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/56/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f6c0e84c7c984b73a6a67497a8df5cd7/index.html : SUCCESS in 16m 23s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/56/check/real-db-upgrade_nova_percona_user_002:th-percona/4ca0786c8acf4e57b5c4734cc42cef59/index.html : SUCCESS in 15m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":56},{"id":"58bfb5ddc6d2f871247d34169726b03fd0f0524a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-04 19:51:24.000000000","message":"Patch Set 56: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/56/21598/index.html : SUCCESS in 01 h 15 m 24 s","accounts_in_message":[],"_revision_number":56},{"id":"bb2da59ddd03eb1fbf8bb9fb41d9de5684500fb1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-04 21:03:21.000000000","message":"Patch Set 56:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/56/ : FAILURE in 1h 24m 42s","accounts_in_message":[],"_revision_number":56},{"id":"a74d79ae168e4ae0dc1bda2cb2d9f0ba280d6a53","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-04 21:39:15.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/56/check/dsvm-tempest-xen/fecb640 : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":56},{"id":"ab1d137c1cc590fcbf5ef58e78ae32dcef48a9fb","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 21:47:42.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"b61e63da49b7dfd37f94f1c4d14a2ecfce277932","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-04 22:20:41.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"6e38e1629ede5d4b68d78f87461188e7224f28ab","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-04 23:09:33.000000000","message":"Patch Set 58:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/58/check/check-ibm-tempest-dsvm-full/4945b8d/ : SUCCESS in 46m 24s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"dc37a723afacd10b86330cfbab19985bc97ee922","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-04 23:25:08.000000000","message":"Patch Set 58: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/58/21644/index.html : SUCCESS in 01 h 01 m 25 s","accounts_in_message":[],"_revision_number":58},{"id":"1023bae1acbe9f095228acd4dbcaae1795d1456a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-05 00:03:24.000000000","message":"Patch Set 58:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/58/ : FAILURE in 1h 14m 24s","accounts_in_message":[],"_revision_number":58},{"id":"dabdeb0252b8fa79b2ccc8dd1134163fdb726f63","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-05 00:29:30.000000000","message":"Patch Set 58: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/58/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/88da333521ea48208136fd97576c9051/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/58/check/real-db-upgrade_nova_mysql_user_001:th-mysql/64d104af7cea429fabdc345c11e3fda7/index.html : SUCCESS in 15m 42s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/58/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/326501b11800453e9e96de4153616850/index.html : SUCCESS in 12m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/58/check/real-db-upgrade_nova_percona_user_001:th-percona/efd2eb85485d44c18a6b02c800fcf653/index.html : SUCCESS in 16m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/58/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a0e576e74c124a96aac46042140c6b45/index.html : SUCCESS in 15m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/58/check/real-db-upgrade_nova_percona_user_002:th-percona/0487c3936d484f3b931fa3b4c2634e08/index.html : SUCCESS in 13m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":58},{"id":"b9f17a03c8f953abb26a337f124268f770f58ddb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-05 00:55:29.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/58/check/gate-nova-pep8/ea85640/ : SUCCESS in 7m 15s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/58/check/gate-nova-docs/1332d0f//doc/build/html/ : SUCCESS in 14m 04s\n- gate-nova-python27 http://logs.openstack.org/35/193135/58/check/gate-nova-python27/a90fe37/ : SUCCESS in 14m 08s\n- gate-nova-python34 http://logs.openstack.org/35/193135/58/check/gate-nova-python34/0c41d7d/ : SUCCESS in 25m 37s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-full/70566d9/ : SUCCESS in 1h 09m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-postgres-full/290892e/ : SUCCESS in 46m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-neutron-full/eb2ec6f/ : SUCCESS in 1h 20m 12s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/58/check/gate-grenade-dsvm/03c2511/ : SUCCESS in 44m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-large-ops/5f36c11/ : SUCCESS in 33m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-neutron-large-ops/94ea019/ : SUCCESS in 30m 31s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/58/check/gate-nova-releasenotes/7544a1a//releasenotes/build/html/ : SUCCESS in 8m 39s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/58/check/gate-nova-tox-functional/ae51d21/ : SUCCESS in 12m 31s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/58/check/gate-grenade-dsvm-multinode/8e7683f/ : SUCCESS in 52m 05s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3b8ff28/ : SUCCESS in 39m 22s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-nova-v20-api/19455a8/ : SUCCESS in 32m 56s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-nova-v20-api-legacy/f7f90b6/ : SUCCESS in 38m 05s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-cells/29bf2f0/ : SUCCESS in 38m 19s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-full-ceph/d3d4b82/ : SUCCESS in 49m 32s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/9a578f6/ : SUCCESS in 50m 17s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/58/check/gate-tempest-dsvm-multinode-full/a2d9285/ : SUCCESS in 52m 41s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/58/check/gate-compute-api-guide/b8e120c//api-guide/build/html/ : SUCCESS in 8m 21s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/58/check/gate-nova-commit-message/9bf4635/ : SUCCESS in 1m 26s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/58/check/gate-nova-python27-constraints/94f4190/ : SUCCESS in 15m 18s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/58/check/gate-nova-python34-constraints/9b02753/ : SUCCESS in 28m 59s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/58/check/gate-nova-docs-constraints/b8b5cfd//doc/build/html/ : SUCCESS in 15m 03s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/58/check/gate-nova-pep8-constraints/54f0f30/ : SUCCESS in 9m 26s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/58/check/gate-nova-tox-constraints-functional-constraints/0ad2f36/ : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"a65ada4a7d21361be9f990460e7fcae87154a607","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-05 01:10:27.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/58/check/dsvm-tempest-xen/ef95a7a : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":58},{"id":"927969822bf22de1bdc0a653d8bac8d7190a88be","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-05 02:48:41.000000000","message":"Patch Set 54:\n\n* pci-test http://52.27.155.124/pci/193135/54 : FAILURE","accounts_in_message":[],"_revision_number":54},{"id":"8d9abd0752d2f210db66c48e4d6eacd38879b990","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-05 09:44:30.000000000","message":"Patch Set 56:\n\n* pci-test http://52.27.155.124/pci/193135/56 : SUCCESS","accounts_in_message":[],"_revision_number":56},{"id":"4d69d901fb41f4405a9136e8749b856a6ec8ae54","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-05 11:44:37.000000000","message":"Patch Set 54:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/54/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":54},{"id":"eb361e770e08e48de3289b2894b69fdd98dd6a5c","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-01-05 12:13:02.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"f61f0ccacb32675db958c9f2b13354e96c8c03a0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-05 13:13:27.000000000","message":"Patch Set 57:\n\n* pci-test http://52.27.155.124/pci/193135/57 : SUCCESS","accounts_in_message":[],"_revision_number":57},{"id":"b63695704307d36e248fb510b40783b3cddee3fe","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-05 13:39:33.000000000","message":"Patch Set 58:\n\n* pci-test http://52.27.155.124/pci/193135/58 : SUCCESS","accounts_in_message":[],"_revision_number":58},{"id":"201874e8d8dfc2d6288a28af8175e82d9e861d26","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-05 14:31:46.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.","accounts_in_message":[],"_revision_number":59},{"id":"fb92dc8177800768dae2046f49bbc550465137d1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-05 14:56:13.000000000","message":"Patch Set 59: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/59/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/560df079578e4cb68330ae98a72b6d1b/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/59/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3ba77875ac4c4def82cafe53cc3f5b81/index.html : SUCCESS in 12m 49s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/59/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3dd11c790d974178b4ff051a326d3adc/index.html : SUCCESS in 14m 38s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/59/check/real-db-upgrade_nova_percona_user_001:th-percona/5142055586dd480682dd6179d2a433a4/index.html : SUCCESS in 16m 42s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/59/check/real-db-upgrade_nova_mysql_user_002:th-mysql/8cde8b9d98714e7e94f5a0b8efdf9a5a/index.html : SUCCESS in 14m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/59/check/real-db-upgrade_nova_percona_user_002:th-percona/21bcc86d006f43038caaf987e24d2d45/index.html : SUCCESS in 14m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":59},{"id":"a77ec73f0f9592b153aba8dc200d854de15dc88e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-05 15:25:52.000000000","message":"Patch Set 59: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/59/21751/index.html : SUCCESS in 50 m 25 s","accounts_in_message":[],"_revision_number":59},{"id":"17dcb587e550d2ad64c37ec25e7a5a303034a6ff","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-05 15:27:50.000000000","message":"Patch Set 59:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/59/check/check-ibm-tempest-dsvm-full/39981e9/ : SUCCESS in 54m 41s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"877b1beda0048a0c6050677c5d2323ac2e47146a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-05 15:31:03.000000000","message":"Patch Set 59:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/59/ : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":59},{"id":"fe44e233bc2055a5915147cc9c303d3c987dd63c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-05 15:47:45.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/59/check/dsvm-tempest-xen/cbcd5a2 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":59},{"id":"4362b3f2196eea092e8a4ef5ce6958fbc2ee0ab2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-05 16:08:23.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/59/check/gate-nova-pep8/f39551f/ : SUCCESS in 8m 07s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/59/check/gate-nova-docs/886beb1//doc/build/html/ : SUCCESS in 11m 46s\n- gate-nova-python27 http://logs.openstack.org/35/193135/59/check/gate-nova-python27/54ead59/ : SUCCESS in 11m 10s\n- gate-nova-python34 http://logs.openstack.org/35/193135/59/check/gate-nova-python34/20824c3/ : SUCCESS in 26m 28s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-full/811ea59/ : FAILURE in 57m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-postgres-full/c2fdba4/ : SUCCESS in 49m 14s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-neutron-full/48c871b/ : SUCCESS in 1h 09m 37s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/59/check/gate-grenade-dsvm/7ee8da5/ : SUCCESS in 52m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-large-ops/5fc9d50/ : SUCCESS in 33m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-neutron-large-ops/fd5d469/ : SUCCESS in 33m 23s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/59/check/gate-nova-releasenotes/891e1fe//releasenotes/build/html/ : SUCCESS in 7m 35s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/59/check/gate-nova-tox-functional/f80cd7c/ : SUCCESS in 12m 50s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/59/check/gate-grenade-dsvm-multinode/483f26c/ : SUCCESS in 58m 28s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b6ad66e/ : SUCCESS in 40m 45s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-nova-v20-api/118278e/ : SUCCESS in 47m 04s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-nova-v20-api-legacy/32e71c9/ : SUCCESS in 43m 15s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-cells/7120357/ : SUCCESS in 44m 01s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-full-ceph/1a69f85/ : SUCCESS in 48m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/aa156d9/ : FAILURE in 56m 32s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/59/check/gate-tempest-dsvm-multinode-full/6b7e602/ : FAILURE in 1h 16m 51s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/59/check/gate-compute-api-guide/109eaf4//api-guide/build/html/ : SUCCESS in 7m 02s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/59/check/gate-nova-commit-message/d95bc88/ : SUCCESS in 1m 49s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/59/check/gate-nova-python27-constraints/d3c7234/ : SUCCESS in 12m 01s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/59/check/gate-nova-python34-constraints/5ba8c4e/ : SUCCESS in 25m 57s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/59/check/gate-nova-docs-constraints/a1daef2//doc/build/html/ : SUCCESS in 10m 05s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/59/check/gate-nova-pep8-constraints/a675dbc/ : SUCCESS in 8m 16s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/59/check/gate-nova-tox-constraints-functional-constraints/4408d51/ : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"17c4cde55ea79471a1e012a8249f6f87811331c6","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2016-01-05 18:17:29.000000000","message":"Patch Set 59: Code-Review-1\n\nI don\u0027t see how this patch addresses detaching in case more than one isntance using the volume is on the same host","accounts_in_message":[],"_revision_number":59},{"id":"533f8bb0cb2999b127c70067f23c983401f56195","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-05 21:52:08.000000000","message":"Patch Set 54:\n\n* pci-test http://52.27.155.124/pci/193135/54 : SUCCESS","accounts_in_message":[],"_revision_number":54},{"id":"014fed668a48d6307b2b43c29405e58f1445b9ab","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-05 23:22:05.000000000","message":"Uploaded patch set 60: Patch Set 59 was rebased.","accounts_in_message":[],"_revision_number":60},{"id":"b2570aaa1c5c7b145011665f12af91df76c4f3c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-05 23:23:07.000000000","message":"Patch Set 60:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":60},{"id":"41fcea6f7f70ce35cc8f1f4485e044db2d579f0b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-05 23:23:10.000000000","message":"Patch Set 60:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":60},{"id":"c643390b86c0e4ee222a5483970f1a41e8097f35","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-05 23:23:32.000000000","message":"Patch Set 60: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":60},{"id":"560c5dcca63284912d6741654b0e9350de3fc743","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-05 23:23:36.000000000","message":"Patch Set 60:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":60},{"id":"3a749f145baddbb22f234d2dc19c9d419ce41fc7","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-05 23:25:06.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.","accounts_in_message":[],"_revision_number":61},{"id":"0be3961f2f96837af0a830bded7e60deca65de0a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-05 23:25:58.000000000","message":"Patch Set 61: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":61},{"id":"fdf51d286c672af5352f2da44372bfdce431555d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-05 23:26:20.000000000","message":"Patch Set 61:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":61},{"id":"eb989d8ac99f82cbf940e096592804ecc77d8868","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-05 23:28:36.000000000","message":"Patch Set 61:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":61},{"id":"c1118f36d4ab6d62e8f9c83c3622efe43c45cb8f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-05 23:28:48.000000000","message":"Patch Set 61:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":61},{"id":"0923570771e79695767284b1af0cc1428950d13a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-05 23:30:20.000000000","message":"Patch Set 40: Verified+1\n\nBuild successful. \n\n[http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/40/771 nova-vc6-dsvm-tempest] : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"56449f42439e0081d3cccf0e15ce7c4dc3c1304c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 00:08:22.000000000","message":"Uploaded patch set 63: Patch Set 61 was rebased.","accounts_in_message":[],"_revision_number":63},{"id":"1b1f963825923cf603221196986d26b0cb38fdb5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-06 00:09:28.000000000","message":"Patch Set 63:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":63},{"id":"b05b6f73e4de7fbea5422430f54a5f7709d87e41","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-06 00:09:37.000000000","message":"Patch Set 63: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":63},{"id":"d9caf1bf9468cf52efdf05b38500858f7ea65a0f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-06 00:09:38.000000000","message":"Patch Set 63:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":63},{"id":"bb2aa851c3bffffe13cc8e0ff18de2e6f76d687b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-06 00:10:04.000000000","message":"Patch Set 63:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":63},{"id":"063232d3d92e5536ac278bfc9410b8ca9d510f2e","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 00:10:16.000000000","message":"Uploaded patch set 64: Patch Set 63 was rebased.","accounts_in_message":[],"_revision_number":64},{"id":"0c407e34d090f787676c3e9aedb63a810249df45","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-06 00:12:37.000000000","message":"Patch Set 64: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":64},{"id":"d8ef68d4fda5e573d9e671c9f3d8e5788468d127","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-06 00:12:51.000000000","message":"Patch Set 64:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":64},{"id":"4c6e745e9eccad65c416e632d656a4bb7fbc26c0","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 00:13:28.000000000","message":"Uploaded patch set 65: Patch Set 64 was rebased.","accounts_in_message":[],"_revision_number":65},{"id":"7104734904b4b1167db56c27c27887f4072bb646","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-06 00:14:17.000000000","message":"Patch Set 65: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":65},{"id":"cabaa874c6636a994a7584441b8a92ff72abe799","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-06 00:14:50.000000000","message":"Patch Set 65:\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":65},{"id":"c2b6d3e1d2e8125dbf38758eec6b942fa34d503b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-06 00:17:54.000000000","message":"Patch Set 65:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":65},{"id":"1012d24aa2674e09326d08b7f9bbb47ec3043278","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-06 00:18:06.000000000","message":"Patch Set 65:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":65},{"id":"161a8e77ef15618b19eddf2a353252573f817f0a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 01:12:44.000000000","message":"Patch Set 59:\n\n* pci-test http://52.27.155.124/pci/193135/59 : SUCCESS","accounts_in_message":[],"_revision_number":59},{"id":"e1dcc4f5afb6b815c7fdbb2e1a8d0b3758511945","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-06 01:19:49.000000000","message":"Patch Set 65: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/65/21840/index.html : SUCCESS in 58 m 53 s","accounts_in_message":[],"_revision_number":65},{"id":"c9900e714d603d0464bb9d2f4dd3982b385674be","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-06 06:03:07.000000000","message":"Patch Set 41:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/41/787 : FAILURE","accounts_in_message":[],"_revision_number":41},{"id":"ef1fce4c46fd23b4a11515e255de0544ee816cbc","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 06:45:48.000000000","message":"Patch Set 60:\n\n* pci-test http://52.27.155.124/pci/193135/60 : SUCCESS","accounts_in_message":[],"_revision_number":60},{"id":"81e63a037ae6c87ea86a5fcc8ed30e0e6282c459","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 07:14:29.000000000","message":"Patch Set 61:\n\n* pci-test http://52.27.155.124/pci/193135/61 : SUCCESS","accounts_in_message":[],"_revision_number":61},{"id":"173cbb285810f4163689686f5325c91774661918","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 07:32:38.000000000","message":"Patch Set 63:\n\n* pci-test http://52.27.155.124/pci/193135/63 : SUCCESS","accounts_in_message":[],"_revision_number":63},{"id":"9b9dc1f8202096d9116b25352780b57da91d16bd","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 07:58:33.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.","accounts_in_message":[],"_revision_number":66},{"id":"1617972a876e2dd9f4a9d7ce8031c5536500e0d3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-06 08:05:01.000000000","message":"Patch Set 57:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/57/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":57},{"id":"e5c3a7c87dee0f8c7733ac29267d7c99c5dda72d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 08:09:27.000000000","message":"Patch Set 64:\n\n* pci-test http://52.27.155.124/pci/193135/64 : SUCCESS","accounts_in_message":[],"_revision_number":64},{"id":"c7ebafee303f4505db09cb68859f65e522ab63b6","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-06 08:22:33.000000000","message":"Patch Set 66: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/66/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2820cf9348df4e4f945594ddf50c6d34/index.html : SUCCESS in 14m 52s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/66/check/real-db-upgrade_nova_mysql_user_001:th-mysql/28ae45d4cd1b416db56315f68a1b8e0d/index.html : SUCCESS in 15m 44s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/66/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/34bc92791fc54945aa8da4ee8663ed07/index.html : SUCCESS in 12m 52s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/66/check/real-db-upgrade_nova_percona_user_001:th-percona/7825350802d04065ac3cdae35ef5362f/index.html : SUCCESS in 16m 33s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/66/check/real-db-upgrade_nova_mysql_user_002:th-mysql/369426aefacc492bb7328faa53508b9f/index.html : SUCCESS in 12m 59s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/66/check/real-db-upgrade_nova_percona_user_002:th-percona/9a9a9f6143384a518090ea7582a129c6/index.html : SUCCESS in 14m 29s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":66},{"id":"c581359c0c5994095dbdb70b6eddce50e306c668","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 08:27:46.000000000","message":"Patch Set 65:\n\n* pci-test http://52.27.155.124/pci/193135/65 : SUCCESS","accounts_in_message":[],"_revision_number":65},{"id":"75dad4127fd5b803d324b5dd94ea85ebe93a252f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-06 08:45:42.000000000","message":"Patch Set 58:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/58/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":58},{"id":"fadba34ff25e877efd35aebb9e939ef6be66a564","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-06 08:54:46.000000000","message":"Patch Set 66:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/66/check/check-ibm-tempest-dsvm-full/3092d53/ : SUCCESS in 54m 39s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"c3e890352cf55023668dadad8160cbf063a9c444","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-06 09:09:57.000000000","message":"Patch Set 66: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/66/21897/index.html : SUCCESS in 01 h 06 m 04 s","accounts_in_message":[],"_revision_number":66},{"id":"2b843658d2d89b3773c9bc5d2b3f1a7d48adc09a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-06 09:20:32.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/66/check/dsvm-tempest-xen/a32a3eb : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":66},{"id":"c96cf314d407d075ae1b3af4de1dcec3aba398a3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-06 09:27:03.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/66/check/gate-nova-pep8/ec953ed/ : SUCCESS in 8m 07s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/66/check/gate-nova-docs/5410ec3//doc/build/html/ : SUCCESS in 11m 31s\n- gate-nova-python27 http://logs.openstack.org/35/193135/66/check/gate-nova-python27/2eefd2c/ : SUCCESS in 15m 59s\n- gate-nova-python34 http://logs.openstack.org/35/193135/66/check/gate-nova-python34/2249c83/ : SUCCESS in 24m 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-full/0831fe3/ : SUCCESS in 41m 18s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-postgres-full/172428a/ : SUCCESS in 38m 13s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-neutron-full/a458a4a/ : SUCCESS in 1h 10m 41s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/66/check/gate-grenade-dsvm/7698914/ : SUCCESS in 44m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-large-ops/6e5ef39/ : SUCCESS in 38m 36s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-neutron-large-ops/977c45e/ : SUCCESS in 32m 34s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/66/check/gate-nova-releasenotes/c4eeb4b//releasenotes/build/html/ : SUCCESS in 6m 29s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/66/check/gate-nova-tox-functional/e645e87/ : SUCCESS in 11m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/66/check/gate-grenade-dsvm-multinode/92b79ee/ : SUCCESS in 53m 42s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/42d1a4f/ : SUCCESS in 35m 14s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-nova-v20-api/5c1dba8/ : SUCCESS in 37m 26s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-nova-v20-api-legacy/02f4d7e/ : SUCCESS in 53m 33s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-cells/e4140fb/ : SUCCESS in 40m 33s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-full-ceph/5bee161/ : SUCCESS in 47m 43s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/09c289b/ : SUCCESS in 48m 07s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/66/check/gate-tempest-dsvm-multinode-full/58f84bc/ : FAILURE in 1h 11m 30s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/66/check/gate-compute-api-guide/ad1403c//api-guide/build/html/ : SUCCESS in 6m 33s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/66/check/gate-nova-commit-message/1c5aa0d/ : SUCCESS in 1m 24s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/66/check/gate-nova-python27-constraints/1227bf6/ : SUCCESS in 15m 18s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/66/check/gate-nova-python34-constraints/fa61bdc/ : SUCCESS in 26m 13s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/66/check/gate-nova-docs-constraints/9abecf9//doc/build/html/ : SUCCESS in 11m 57s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/66/check/gate-nova-pep8-constraints/9781008/ : SUCCESS in 7m 54s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/66/check/gate-nova-tox-constraints-functional-constraints/e212e9d/ : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"07058a4bafc327554e5e33b4f97ab82c26909aa5","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 09:32:37.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"eaaa9f1b8d0f83a8b9b95b7291cbffc0ab9ed8ab","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 09:39:06.000000000","message":"Patch Set 66:\n\nHi Nikola,\n\nThanks for the review.\n\nBy your concern do you mean that disconnecting a volume from one instance would remove all the initiators for the target on that host?\n\nThanks,\nIldiko","accounts_in_message":[],"_revision_number":66},{"id":"4522f358ced0afd84a59c63042f1279df93c9349","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-06 09:39:59.000000000","message":"Patch Set 66:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/66/ : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":66},{"id":"dfaed30de8fdcf535228a2528e961539028ecdac","author":{"_account_id":13002,"name":"Shyama","email":"shyvenug@in.ibm.com","username":"shyama"},"date":"2016-01-06 10:58:14.000000000","message":"Patch Set 66:\n\nIldiko, I agree with Nikola the fix in cinder https://review.openstack.org/#/c/255595/ takes care of preventing export volume from being called when there are more \u0027attached host\u0027 using the volume. In many cinder drivers the terminate_connection implementation removes the cinder volume connection and remove_export does nothing.","accounts_in_message":[],"_revision_number":66},{"id":"ada9abb93c9b902f066ea047492ba979312142d3","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-06 11:03:53.000000000","message":"Patch Set 66:\n\nHi Shyama,\n\nDo you mean that it\u0027s removed on Cinder side based on how terminate_connection is implemented and in this sense don\u0027t call terminate_connection in certain cases?\n\nThanks,\nIldiko","accounts_in_message":[],"_revision_number":66},{"id":"e72588d7b8ea26cc83997466951c72c8fbfa5675","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-06 11:23:37.000000000","message":"Patch Set 66:\n\n* pci-test http://52.27.155.124/pci/193135/66 : SUCCESS","accounts_in_message":[],"_revision_number":66},{"id":"73589c53668d41182251db05822d754aa056723a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-06 12:49:59.000000000","message":"Patch Set 42:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/42/802 : FAILURE","accounts_in_message":[],"_revision_number":42},{"id":"895a346b3d0fafcf7740c6e9988b09cca82af615","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-06 16:38:22.000000000","message":"Patch Set 43:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/43/808 : FAILURE","accounts_in_message":[],"_revision_number":43},{"id":"6c2f75128fbbd231a417d1290b0db11a492d017b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-06 20:15:27.000000000","message":"Patch Set 44: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/44/814 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"850ac0318f9b05a9b06e4059d46d7ff14d4514b7","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-07 10:53:10.000000000","message":"Uploaded patch set 67.","accounts_in_message":[],"_revision_number":67},{"id":"01550696b4ff2b8880fd05f20fb80037d309e8ed","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-07 11:11:13.000000000","message":"Patch Set 67:\n\n* pci-test http://52.27.155.124/pci/193135/67 : SUCCESS","accounts_in_message":[],"_revision_number":67},{"id":"3349f5eac2ee106da3c9d47412c8a4292f147474","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-07 11:17:51.000000000","message":"Patch Set 67: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/67/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/74514b1f34f34ceea4a3f4c3809bb321/index.html : SUCCESS in 14m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/67/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6135c1baf4bb480f831850d369c96a79/index.html : SUCCESS in 17m 17s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/67/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e873b20778a0443aa473627739641cc6/index.html : SUCCESS in 15m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/67/check/real-db-upgrade_nova_percona_user_001:th-percona/d77febcf08f94fe0a209c94200f6f63c/index.html : SUCCESS in 16m 20s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/67/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0e9cb917b7194df3b1de900cb4691905/index.html : SUCCESS in 15m 10s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/67/check/real-db-upgrade_nova_percona_user_002:th-percona/3498882a6da947b392f2049a54c658a8/index.html : SUCCESS in 15m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":67},{"id":"c607357e9d308644e9d5b30b6558c6ecd68f0148","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-07 11:48:12.000000000","message":"Patch Set 67:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/67/check/check-ibm-tempest-dsvm-full/90825e7/ : SUCCESS in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"634db6c7f061860519e6f088eaf0fb96e240cd15","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-07 11:56:09.000000000","message":"Patch Set 67: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/67/22123/index.html : FAILURE in 01 h 00 m 10 s","accounts_in_message":[],"_revision_number":67},{"id":"6e75cde681fadb52de6bc093f00b5a911c992e48","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-07 12:08:34.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/67/check/gate-nova-pep8/f10d55b/ : SUCCESS in 7m 22s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/67/check/gate-nova-docs/f347543//doc/build/html/ : SUCCESS in 16m 33s\n- gate-nova-python27 http://logs.openstack.org/35/193135/67/check/gate-nova-python27/fb1b1a1/ : SUCCESS in 12m 04s\n- gate-nova-python34 http://logs.openstack.org/35/193135/67/check/gate-nova-python34/5517f53/ : SUCCESS in 25m 48s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-full/bde8e54/ : SUCCESS in 44m 42s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-postgres-full/ab97265/ : SUCCESS in 39m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-neutron-full/2c8ec60/ : SUCCESS in 1h 09m 39s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/67/check/gate-grenade-dsvm/fd2ec50/ : SUCCESS in 51m 28s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-large-ops/5545b57/ : SUCCESS in 46m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-neutron-large-ops/849fdf8/ : SUCCESS in 27m 48s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/67/check/gate-nova-releasenotes/2555783//releasenotes/build/html/ : SUCCESS in 9m 05s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/67/check/gate-nova-tox-functional/aeb3ddc/ : SUCCESS in 16m 22s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/67/check/gate-grenade-dsvm-multinode/d980e2d/ : SUCCESS in 1h 10m 28s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/1108084/ : SUCCESS in 36m 50s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-nova-v20-api/25aaaa8/ : SUCCESS in 37m 22s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-nova-v20-api-legacy/a77b967/ : SUCCESS in 33m 59s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-cells/39733c0/ : SUCCESS in 1h 02m 48s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-full-ceph/9cb3577/ : SUCCESS in 46m 57s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4b19a23/ : SUCCESS in 46m 25s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/67/check/gate-tempest-dsvm-multinode-full/3f46df6/ : FAILURE in 56m 25s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/67/check/gate-compute-api-guide/2663152//api-guide/build/html/ : SUCCESS in 7m 06s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/67/check/gate-nova-commit-message/347a1c1/ : SUCCESS in 1m 53s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/67/check/gate-nova-python27-constraints/6579af5/ : SUCCESS in 16m 50s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/67/check/gate-nova-python34-constraints/f29f917/ : SUCCESS in 29m 22s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/67/check/gate-nova-docs-constraints/74c45a9//doc/build/html/ : SUCCESS in 14m 57s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/67/check/gate-nova-pep8-constraints/d31c418/ : SUCCESS in 8m 10s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/67/check/gate-nova-tox-constraints-functional-constraints/b7e6b9f/ : SUCCESS in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"41c7b837c382608bad344e2027a1f7c9376092c5","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-07 12:09:28.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/67/check/dsvm-tempest-xen/b3103eb : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":67},{"id":"fae8e633e2de0117c09250d1f7b49e911a258d33","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-07 13:14:11.000000000","message":"Patch Set 67:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/67/ : FAILURE in 2h 19m 35s","accounts_in_message":[],"_revision_number":67},{"id":"f9cb23e687f91391c7019cb0ee297ce34ec473f1","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-07 17:53:07.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.","accounts_in_message":[],"_revision_number":68},{"id":"09b0eb376b1843e8d69bef782b2337b5192f7ff7","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-07 17:57:02.000000000","message":"Uploaded patch set 69: Patch Set 68 was rebased.","accounts_in_message":[],"_revision_number":69},{"id":"ccf13d86b158ec1137b00f493849ae863ce12f21","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-07 18:52:46.000000000","message":"Patch Set 69:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/69/check/check-ibm-tempest-dsvm-full/88d8cac/ : SUCCESS in 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"5467eb2d2927b6e3c8034a36424635215aad871f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-07 19:09:19.000000000","message":"Patch Set 69: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/69/22217/index.html : SUCCESS in 56 m 21 s","accounts_in_message":[],"_revision_number":69},{"id":"13c14d2e2e71fd2da9eeedb4eca9bf34863dc4cd","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-07 19:29:01.000000000","message":"Patch Set 68:\n\n* pci-test http://52.27.155.124/pci/193135/68 : SUCCESS","accounts_in_message":[],"_revision_number":68},{"id":"c940ab5d3808089508b0ada4c5561a699974a828","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-07 19:31:59.000000000","message":"Patch Set 69: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/69/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e1578a773bc34b968e6cf374a8e2b75f/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/69/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1f79963391604552a3061dd300e384cf/index.html : SUCCESS in 14m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/69/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/38c0e28b1b0b4581997051d251cb127f/index.html : SUCCESS in 14m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/69/check/real-db-upgrade_nova_percona_user_001:th-percona/50151c5746244afe9230ebc6d2346812/index.html : SUCCESS in 19m 05s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/69/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2d0f016473424ca281a229082afdd5eb/index.html : SUCCESS in 13m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/69/check/real-db-upgrade_nova_percona_user_002:th-percona/bbab2456d1fc4332a6fd9e1d11ada968/index.html : SUCCESS in 12m 52s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":69},{"id":"bf345f40fd1e18e5783cff4e503e28f5b3221432","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-07 19:40:04.000000000","message":"Patch Set 69: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/69/check/gate-nova-pep8/63fb338/ : SUCCESS in 10m 28s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/69/check/gate-nova-docs/5e02c52//doc/build/html/ : SUCCESS in 14m 37s\n- gate-nova-python27 http://logs.openstack.org/35/193135/69/check/gate-nova-python27/a8103d4/ : SUCCESS in 21m 26s\n- gate-nova-python34 http://logs.openstack.org/35/193135/69/check/gate-nova-python34/2c0a186/ : SUCCESS in 29m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-full/baa3c45/ : SUCCESS in 1h 00m 56s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-postgres-full/c22ce93/ : SUCCESS in 51m 14s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-neutron-full/076dee0/ : SUCCESS in 1h 11m 33s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/69/check/gate-grenade-dsvm/2484676/ : SUCCESS in 54m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-large-ops/7063c3e/ : SUCCESS in 26m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-neutron-large-ops/0658446/ : SUCCESS in 34m 07s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/69/check/gate-nova-releasenotes/bdb5507//releasenotes/build/html/ : SUCCESS in 9m 06s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/69/check/gate-nova-tox-functional/3add654/ : SUCCESS in 15m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/69/check/gate-grenade-dsvm-multinode/08c9b31/ : SUCCESS in 51m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/57149a0/ : SUCCESS in 42m 26s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-nova-v20-api/31e503a/ : SUCCESS in 45m 31s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-nova-v20-api-legacy/6f9541b/ : SUCCESS in 34m 00s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-cells/d1621b6/ : SUCCESS in 45m 55s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-full-ceph/65bc13f/ : SUCCESS in 51m 01s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/b2992eb/ : SUCCESS in 40m 43s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/69/check/gate-tempest-dsvm-multinode-full/8a8fe59/ : FAILURE in 1h 22m 47s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/69/check/gate-compute-api-guide/ff8c62d//api-guide/build/html/ : SUCCESS in 8m 39s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/69/check/gate-nova-commit-message/37dbc80/ : SUCCESS in 2m 29s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/69/check/gate-nova-python27-constraints/33cd570/ : SUCCESS in 15m 19s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/69/check/gate-nova-python34-constraints/3b86614/ : SUCCESS in 25m 26s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/69/check/gate-nova-docs-constraints/979891d//doc/build/html/ : SUCCESS in 14m 27s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/69/check/gate-nova-pep8-constraints/506dd92/ : SUCCESS in 9m 16s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/69/check/gate-nova-tox-constraints-functional-constraints/85db189/ : SUCCESS in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"5f2773e12538b80cc52a8b8fd3a61df742aea79e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-07 19:48:55.000000000","message":"Patch Set 69: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/69/check/dsvm-tempest-xen/c8e286e : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":69},{"id":"327922cae34cc2611e7d07603961c7ba93e71699","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-07 19:57:36.000000000","message":"Patch Set 69:\n\n* pci-test http://52.27.155.124/pci/193135/69 : SUCCESS","accounts_in_message":[],"_revision_number":69},{"id":"7364ecb8021955f6a2171ba4e174dcb6d3f11106","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-07 22:34:31.000000000","message":"Patch Set 69:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/69/ : FAILURE in 3h 56m 19s","accounts_in_message":[],"_revision_number":69},{"id":"f212187caf63dbe3e90119ce405d2b2bffa479f8","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-08 17:15:48.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"0ec1b6a1c6aae768ebc1196c0e865c5ad7f3c500","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-08 17:53:20.000000000","message":"Patch Set 70:\n\n* pci-test http://52.27.155.124/pci/193135/70 : SUCCESS","accounts_in_message":[],"_revision_number":70},{"id":"94279ff3c44b1710dee5b912423d3be29f01ec9e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-08 18:12:25.000000000","message":"Patch Set 70:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/70/check/check-ibm-tempest-dsvm-full/8e72cc4/ : SUCCESS in 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"8d49ccad26b311418d0bb2a0693992b48256d80c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-08 18:31:00.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/70/check/dsvm-tempest-xen/9e3739a : FAILURE in 1h 09m 47s","accounts_in_message":[],"_revision_number":70},{"id":"cdacf8aadbd5a08dbbdb356538e7931b63150ac3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-08 18:33:27.000000000","message":"Patch Set 70: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/70/22361/index.html : SUCCESS in 01 h 07 m 59 s","accounts_in_message":[],"_revision_number":70},{"id":"2bb55c1d3a714791c3bc13d5211d79544e20e15f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-08 19:47:39.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/70/check/gate-nova-pep8/4930233/ : SUCCESS in 10m 22s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/70/check/gate-nova-docs/bfe6b0d//doc/build/html/ : SUCCESS in 16m 21s\n- gate-nova-python27 http://logs.openstack.org/35/193135/70/check/gate-nova-python27/3ec22ca/ : SUCCESS in 14m 21s\n- gate-nova-python34 http://logs.openstack.org/35/193135/70/check/gate-nova-python34/c8f7e40/ : SUCCESS in 29m 23s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-full/b0ec374/ : SUCCESS in 42m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-postgres-full/b501556/ : SUCCESS in 1h 01m 12s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-neutron-full/af038ca/ : FAILURE in 1h 16m 10s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/70/check/gate-grenade-dsvm/cda55e0/ : SUCCESS in 55m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-large-ops/24e54f4/ : SUCCESS in 38m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-neutron-large-ops/8076669/ : SUCCESS in 29m 13s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/70/check/gate-nova-releasenotes/0595175//releasenotes/build/html/ : SUCCESS in 9m 47s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/70/check/gate-nova-tox-functional/6cb7a5c/ : SUCCESS in 15m 27s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/70/check/gate-grenade-dsvm-multinode/b98a14d/ : SUCCESS in 59m 52s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/d031d9d/ : FAILURE in 40m 40s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-nova-v20-api/26e7343/ : SUCCESS in 41m 29s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-nova-v20-api-legacy/06fbe4e/ : SUCCESS in 51m 22s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-cells/a39f1a6/ : SUCCESS in 38m 35s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-full-ceph/0d875fb/ : SUCCESS in 48m 27s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/bf2868e/ : SUCCESS in 54m 48s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-multinode-full/f87c08a/ : FAILURE in 1h 10m 50s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/70/check/gate-compute-api-guide/32a31b0//api-guide/build/html/ : SUCCESS in 9m 21s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/70/check/gate-nova-commit-message/50e808d/ : SUCCESS in 1m 57s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-python27-constraints/36a6cf5/ : SUCCESS in 15m 25s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-python34-constraints/d6fdbd0/ : SUCCESS in 25m 55s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/70/check/gate-nova-docs-constraints/e7e5ce1//doc/build/html/ : SUCCESS in 22m 46s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-pep8-constraints/f55fc53/ : SUCCESS in 9m 56s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-tox-constraints-functional-constraints/1ae3fd5/ : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"249dd903245b135d201e5641daa6200eb0f0d403","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-08 19:54:55.000000000","message":"Patch Set 70:\n\nrecheck","accounts_in_message":[],"_revision_number":70},{"id":"89e6010a707c1c2e42c4995a8194b3ec296af41b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-08 20:21:51.000000000","message":"Patch Set 70:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/70/ : FAILURE in 2h 43m 15s","accounts_in_message":[],"_revision_number":70},{"id":"80f5d7dcbad7962a3c0227c366dfd0aa160339ff","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-08 20:49:45.000000000","message":"Patch Set 70:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/70/check/check-ibm-tempest-dsvm-full/edbb553/ : FAILURE in 54m 04s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"06ca9b32cdb8ec7ff88d701db167293ef09ab05b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-08 21:17:30.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/70/check/gate-nova-pep8/980931f/ : SUCCESS in 9m 43s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/70/check/gate-nova-docs/b4d8d8b//doc/build/html/ : SUCCESS in 9m 09s\n- gate-nova-python27 http://logs.openstack.org/35/193135/70/check/gate-nova-python27/e2c12cc/ : SUCCESS in 11m 13s\n- gate-nova-python34 http://logs.openstack.org/35/193135/70/check/gate-nova-python34/29a56f8/ : SUCCESS in 28m 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-full/c6b5de8/ : SUCCESS in 46m 04s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-postgres-full/7964767/ : SUCCESS in 49m 50s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-neutron-full/5a8102a/ : SUCCESS in 1h 03m 24s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/70/check/gate-grenade-dsvm/201e9ec/ : SUCCESS in 50m 32s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-large-ops/334d671/ : SUCCESS in 38m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-neutron-large-ops/0c57f74/ : SUCCESS in 33m 10s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/70/check/gate-nova-releasenotes/7391a9d//releasenotes/build/html/ : SUCCESS in 10m 02s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/70/check/gate-nova-tox-functional/fe91759/ : SUCCESS in 16m 31s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/70/check/gate-grenade-dsvm-multinode/e687239/ : SUCCESS in 1h 06m 32s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ae69bb0/ : SUCCESS in 42m 24s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-nova-v20-api/3a36c80/ : SUCCESS in 35m 34s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-nova-v20-api-legacy/5d9428d/ : SUCCESS in 33m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-cells/f0fbac5/ : SUCCESS in 44m 01s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-full-ceph/f4105c3/ : SUCCESS in 45m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/621635a/ : SUCCESS in 40m 15s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/70/check/gate-tempest-dsvm-multinode-full/a687447/ : FAILURE in 52m 09s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/70/check/gate-compute-api-guide/0d250df//api-guide/build/html/ : SUCCESS in 13m 51s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/70/check/gate-nova-commit-message/94a2683/ : SUCCESS in 1m 31s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-python27-constraints/adc2c1d/ : SUCCESS in 15m 24s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-python34-constraints/7321a98/ : SUCCESS in 28m 30s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/70/check/gate-nova-docs-constraints/626ee3b//doc/build/html/ : SUCCESS in 16m 00s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-pep8-constraints/162264e/ : SUCCESS in 10m 12s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/70/check/gate-nova-tox-constraints-functional-constraints/efede37/ : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"83ff551c271b0e000a0f47b7b7ca1b62d3ab6a80","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-08 22:47:09.000000000","message":"Patch Set 70: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/70/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1e85a7863b8540759957b8b3df637d2b/index.html : SUCCESS in 14m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/70/check/real-db-upgrade_nova_mysql_user_001:th-mysql/725323d55a6b40fbba4cec10f3005932/index.html : SUCCESS in 16m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/70/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f32e9c16eb734e7a8a10b595dac70b14/index.html : SUCCESS in 16m 01s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/70/check/real-db-upgrade_nova_percona_user_001:th-percona/f273f0ad086e461aae4b11137c7f9217/index.html : SUCCESS in 14m 29s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/70/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2cdbd9f28cd94721879e28e8bae7b85c/index.html : SUCCESS in 14m 42s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/70/check/real-db-upgrade_nova_percona_user_002:th-percona/374bc773ff2d4738a5b98d25c10a7cd1/index.html : SUCCESS in 14m 59s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":70},{"id":"6f21a44eea47faa743f3294456204df73689abf5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-09 00:24:35.000000000","message":"Patch Set 70:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/70/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":70},{"id":"a2aa3952272452b6b9c52d4aa7f5285527c50489","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-09 04:32:31.000000000","message":"Patch Set 70:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/70/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":70},{"id":"dbaf90401e5b38def3c3b04bfc0a6c6315386ec5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-09 04:55:30.000000000","message":"Patch Set 70: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"2fc0e6ea95e02855d9c105f80d22857c110d6e34","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-09 08:10:25.000000000","message":"Uploaded patch set 71: Patch Set 70 was rebased.","accounts_in_message":[],"_revision_number":71},{"id":"178919849e09d2a2060eeea601d6395b38e571e3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-09 08:27:59.000000000","message":"Patch Set 71:\n\n* pci-test http://52.27.155.124/pci/193135/71 : SUCCESS","accounts_in_message":[],"_revision_number":71},{"id":"b5e3dac02a0aedf68625b95dd3a115b04e7f70bc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-09 08:32:36.000000000","message":"Patch Set 71: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/71/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7dad3acc2dfc4d5b9b55dc2546fcb7b0/index.html : SUCCESS in 14m 21s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/71/check/real-db-upgrade_nova_mysql_user_001:th-mysql/fa9259d9df644c25bad41be1a44a78e0/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/71/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2c34149a7b68448a92766ce6cab6b570/index.html : SUCCESS in 14m 58s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/71/check/real-db-upgrade_nova_percona_user_001:th-percona/a7109aa521cd483b96e9536ea19431c2/index.html : SUCCESS in 16m 23s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/71/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a5a2a22b611e4ce19b2c45d2196d402c/index.html : SUCCESS in 12m 33s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/71/check/real-db-upgrade_nova_percona_user_002:th-percona/eb1aacd070164f61ba3af567173a6b09/index.html : SUCCESS in 15m 12s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":71},{"id":"2af17ef22f0e0ef357505c485c0aedfdb87d0d2e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-09 09:06:19.000000000","message":"Patch Set 71:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/71/check/check-ibm-tempest-dsvm-full/eca0130/ : SUCCESS in 54m 43s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"68d8c1ba8bffb4b47c8b6bddf998a1767f86c662","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-09 09:15:44.000000000","message":"Patch Set 71: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/71/22414/index.html : SUCCESS in 59 m 41 s","accounts_in_message":[],"_revision_number":71},{"id":"6be493925e5829a8d983ac3ff23007f629125314","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-09 09:16:55.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/71/check/dsvm-tempest-xen/93e7a8b : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":71},{"id":"161a5b7c2b65b309ea15468e548df59f3bfc505f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-09 09:18:07.000000000","message":"Patch Set 71: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/71/check/gate-nova-pep8/b90b046/ : SUCCESS in 9m 26s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/71/check/gate-nova-docs/3497987//doc/build/html/ : SUCCESS in 14m 30s\n- gate-nova-python27 http://logs.openstack.org/35/193135/71/check/gate-nova-python27/418fe0e/ : SUCCESS in 14m 47s\n- gate-nova-python34 http://logs.openstack.org/35/193135/71/check/gate-nova-python34/86b359f/ : SUCCESS in 29m 43s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-full/795320a/ : SUCCESS in 50m 47s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-postgres-full/eb5503d/ : FAILURE in 52m 53s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-neutron-full/5c8ddb7/ : SUCCESS in 1h 02m 25s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/71/check/gate-grenade-dsvm/a201480/ : SUCCESS in 47m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-large-ops/c40cdb8/ : SUCCESS in 33m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-neutron-large-ops/a87cd94/ : SUCCESS in 31m 20s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/71/check/gate-nova-releasenotes/fd7042f//releasenotes/build/html/ : SUCCESS in 9m 27s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/71/check/gate-nova-tox-functional/538a9ba/ : SUCCESS in 15m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/71/check/gate-grenade-dsvm-multinode/742946c/ : SUCCESS in 55m 16s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/4d16039/ : FAILURE in 42m 29s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-nova-v20-api/a335d70/ : SUCCESS in 47m 19s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-nova-v20-api-legacy/0021dbf/ : SUCCESS in 47m 02s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-cells/f01e270/ : SUCCESS in 42m 51s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-full-ceph/2de4e93/ : FAILURE in 58m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/48ac3b4/ : SUCCESS in 54m 20s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-multinode-full/dede1f4/ : FAILURE in 1h 00m 33s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/71/check/gate-compute-api-guide/7904d6f//api-guide/build/html/ : SUCCESS in 9m 22s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/71/check/gate-nova-commit-message/b81ce7c/ : SUCCESS in 1m 43s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-python27-constraints/c05c2fb/ : SUCCESS in 15m 54s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-python34-constraints/91d1f2d/ : SUCCESS in 29m 23s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/71/check/gate-nova-docs-constraints/7447378//doc/build/html/ : SUCCESS in 14m 45s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-pep8-constraints/1228bbf/ : SUCCESS in 10m 21s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-tox-constraints-functional-constraints/f768313/ : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"0f2c1ccaa26f74ecdbd68a79e7504f0cea41de69","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-01-09 09:18:32.000000000","message":"Patch Set 71:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full-ceph: https://bugs.launchpad.net/bugs/1475333\n- gate-tempest-dsvm-postgres-full: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":71},{"id":"47bb2fc022892b0d980fa2c3015fba78f0e55a24","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-09 11:29:41.000000000","message":"Patch Set 71:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/71/ : FAILURE in 3h 17m 46s","accounts_in_message":[],"_revision_number":71},{"id":"2183a86449741d527a7087610102487df1d4dd4d","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-09 20:07:08.000000000","message":"Patch Set 71:\n\nrecheck","accounts_in_message":[],"_revision_number":71},{"id":"a3e096a1b781ff9872c5dc587c1959bf2b2d8a05","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-09 21:03:38.000000000","message":"Patch Set 71:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/71/check/check-ibm-tempest-dsvm-full/6cc68a7/ : SUCCESS in 55m 43s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"10f8645e7db6c3969339108937c3463788e28897","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-09 21:18:04.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/71/check/gate-nova-pep8/f2b7294/ : SUCCESS in 9m 10s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/71/check/gate-nova-docs/d2e0d2d//doc/build/html/ : SUCCESS in 15m 07s\n- gate-nova-python27 http://logs.openstack.org/35/193135/71/check/gate-nova-python27/ee7b588/ : SUCCESS in 12m 02s\n- gate-nova-python34 http://logs.openstack.org/35/193135/71/check/gate-nova-python34/0f691e6/ : SUCCESS in 26m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-full/76ab2a5/ : SUCCESS in 40m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-postgres-full/3484022/ : SUCCESS in 48m 21s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-neutron-full/cfc98ea/ : SUCCESS in 1h 01m 01s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/71/check/gate-grenade-dsvm/e892927/ : SUCCESS in 40m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-large-ops/3601555/ : SUCCESS in 29m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-neutron-large-ops/103075b/ : SUCCESS in 27m 30s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/71/check/gate-nova-releasenotes/6d7f12b//releasenotes/build/html/ : SUCCESS in 8m 48s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/71/check/gate-nova-tox-functional/0771c5a/ : SUCCESS in 12m 25s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/71/check/gate-grenade-dsvm-multinode/d5c4468/ : SUCCESS in 52m 48s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2b9ddf1/ : SUCCESS in 39m 33s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-nova-v20-api/6d195f0/ : SUCCESS in 40m 15s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-nova-v20-api-legacy/1ef7224/ : SUCCESS in 41m 14s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-cells/05cebec/ : SUCCESS in 42m 41s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-full-ceph/ccc3fbb/ : SUCCESS in 43m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/2959f46/ : SUCCESS in 45m 25s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/71/check/gate-tempest-dsvm-multinode-full/339d24b/ : FAILURE in 1h 04m 27s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/71/check/gate-compute-api-guide/456b78a//api-guide/build/html/ : SUCCESS in 7m 22s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/71/check/gate-nova-commit-message/a85dd94/ : SUCCESS in 1m 32s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-python27-constraints/0466769/ : SUCCESS in 14m 49s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-python34-constraints/cb23e30/ : SUCCESS in 25m 58s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/71/check/gate-nova-docs-constraints/2db8f57//doc/build/html/ : SUCCESS in 11m 54s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-pep8-constraints/a189122/ : SUCCESS in 8m 20s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/71/check/gate-nova-tox-constraints-functional-constraints/1658306/ : SUCCESS in 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"3ece4a2531038b13f04945e5e22ddfa57e878fce","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-09 23:47:52.000000000","message":"Patch Set 71:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/71/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":71},{"id":"55b2added1f7331cff14a8c452ba89237807ef65","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-11 18:43:42.000000000","message":"Patch Set 59:\n\n@Nikola: From the Cinder drivers that supports multiattach only the lvm driver is problematic on that front you raised. The Cinder guys are working on a fix right now, another option can be to disable the multiattach support for that driver. Otherwise Cinder handles this issue and Nova does not have to be modified.\n\nBest Regards,\nIldiko","accounts_in_message":[],"_revision_number":59},{"id":"69faadbe5b2b7d00993c376204ddbcf676da6d12","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-12 07:44:31.000000000","message":"Patch Set 71: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/71/1074 : SUCCESS","accounts_in_message":[],"_revision_number":71},{"id":"d745870473fb5c3f9ad6f54ec3552376828b313f","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-12 10:05:36.000000000","message":"Uploaded patch set 72: Patch Set 71 was rebased.","accounts_in_message":[],"_revision_number":72},{"id":"935c4508a7593022e11d461749e3e59251e82b4d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-12 10:31:49.000000000","message":"Patch Set 72: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/72/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0f51c6e1193047c3b2d327ba860e9ae4/index.html : SUCCESS in 14m 31s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/72/check/real-db-upgrade_nova_mysql_user_001:th-mysql/96b74fc261404a198e82daa9c7b535ff/index.html : SUCCESS in 15m 55s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/72/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f35599bca5664735be2f11bb86fbe1b7/index.html : SUCCESS in 14m 32s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/72/check/real-db-upgrade_nova_percona_user_001:th-percona/7f1ebf0e0c81409f84d7f0cf4e785755/index.html : SUCCESS in 16m 58s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/72/check/real-db-upgrade_nova_mysql_user_002:th-mysql/23ccfcf210624364a46993630492e0d6/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/72/check/real-db-upgrade_nova_percona_user_002:th-percona/f89ccb12195741749a20c83669864eae/index.html : SUCCESS in 15m 32s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":72},{"id":"738eb8b1a9b5d4f65ff97c5d5174b78c1a37dcc6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-12 10:44:35.000000000","message":"Patch Set 72:\n\n* pci-test http://52.27.155.124/pci/193135/72 : SUCCESS","accounts_in_message":[],"_revision_number":72},{"id":"a0a75dc7956be0da6961e7eff02c0fce0e4ddf9e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-12 11:00:33.000000000","message":"Patch Set 72:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/72/check/check-ibm-tempest-dsvm-full/6cea4f4/ : SUCCESS in 53m 51s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"a2811340de76bd2afb9d6e5707d9a982b07d3cf7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-12 11:11:03.000000000","message":"Patch Set 72: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/72/22718/index.html : SUCCESS in 58 m 52 s","accounts_in_message":[],"_revision_number":72},{"id":"26550abf36fa36305fb8b3a8a231d11179d201f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-12 12:00:44.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/72/check/gate-nova-pep8/5b550a5/ : SUCCESS in 9m 34s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/72/check/gate-nova-docs/b95565e//doc/build/html/ : SUCCESS in 12m 42s\n- gate-nova-python27 http://logs.openstack.org/35/193135/72/check/gate-nova-python27/6de5350/ : SUCCESS in 16m 45s\n- gate-nova-python34 http://logs.openstack.org/35/193135/72/check/gate-nova-python34/e167a88/ : SUCCESS in 28m 52s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-full/6eb07fe/ : SUCCESS in 53m 34s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-postgres-full/56422f1/ : SUCCESS in 49m 21s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-neutron-full/2970f9a/ : SUCCESS in 56m 43s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/72/check/gate-grenade-dsvm/0571211/ : SUCCESS in 56m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-large-ops/5d53834/ : SUCCESS in 39m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-neutron-large-ops/e64cda7/ : SUCCESS in 25m 59s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/72/check/gate-nova-releasenotes/73ea912//releasenotes/build/html/ : SUCCESS in 9m 38s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/72/check/gate-nova-tox-functional/84ed61f/ : SUCCESS in 13m 45s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/72/check/gate-grenade-dsvm-multinode/e2f35e8/ : SUCCESS in 1h 17m 27s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ea61cc1/ : FAILURE in 37m 02s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-nova-v20-api/7966a0b/ : SUCCESS in 35m 10s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-nova-v20-api-legacy/53a224a/ : SUCCESS in 57m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-cells/dbfdd4b/ : SUCCESS in 37m 27s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-full-ceph/8ef6f9e/ : SUCCESS in 44m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/8686969/ : SUCCESS in 47m 45s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/72/check/gate-tempest-dsvm-multinode-full/7b5ee1d/ : FAILURE in 1h 17m 45s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/72/check/gate-compute-api-guide/d8a7496//api-guide/build/html/ : SUCCESS in 7m 15s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/72/check/gate-nova-commit-message/37c8811/ : SUCCESS in 1m 43s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/72/check/gate-nova-python27-constraints/800c39c/ : SUCCESS in 15m 14s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/72/check/gate-nova-python34-constraints/faae9f5/ : SUCCESS in 28m 37s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/72/check/gate-nova-docs-constraints/22359b3//doc/build/html/ : SUCCESS in 11m 21s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/72/check/gate-nova-pep8-constraints/9ce2e9a/ : SUCCESS in 9m 59s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/72/check/gate-nova-tox-constraints-functional-constraints/3fe343e/ : SUCCESS in 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"c257455a3c93a5e0daab091b9405bb6fea58f275","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-12 12:11:31.000000000","message":"Patch Set 72:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/72/ : FAILURE in 2h 04m 06s","accounts_in_message":[],"_revision_number":72},{"id":"6d10f9463dfa2984d5f08134194893ffbc94ec93","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-12 12:17:24.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/72/check/dsvm-tempest-xen/3f57603 : SUCCESS in 2h 10m 50s","accounts_in_message":[],"_revision_number":72},{"id":"f6550c1170b9896a77007a5936a97e7606c62ec1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-12 12:48:41.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/72/check/check-dsvm-tempest-vz7-exe-minimal/81306d1 : SUCCESS in 56m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":72},{"id":"2207a5036afac2b6e63ce38c0f720e23e5dd7371","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-12 14:14:36.000000000","message":"Patch Set 72:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/72/Nova-ML2-Sriov : SUCCESS in 32m 22s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":72},{"id":"d70f5648961e4dd50dde2a29b508cb496404f6ab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-12 15:45:50.000000000","message":"Patch Set 72: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":72},{"id":"f5db6d0609cd258ee75fcaa9a7e50f80513af386","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-12 16:37:23.000000000","message":"Uploaded patch set 73: Patch Set 72 was rebased.","accounts_in_message":[],"_revision_number":73},{"id":"e46d796a30ac8987b1676f4bb9bb951f3172015b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-12 17:08:23.000000000","message":"Patch Set 73: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/73/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bf1a1615f6b54ac7baaf12027f5a8073/index.html : SUCCESS in 15m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/73/check/real-db-upgrade_nova_mysql_user_001:th-mysql/61c5bc4163674afc90f0161f97788af5/index.html : SUCCESS in 15m 13s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/73/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4bb59b4610104374b8338100490a5a2e/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/73/check/real-db-upgrade_nova_percona_user_001:th-percona/b2eebcb65973497989afc0ce330eac5c/index.html : SUCCESS in 14m 07s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/73/check/real-db-upgrade_nova_mysql_user_002:th-mysql/954765907b0c44dc9ddbee9b6d3e7ee6/index.html : SUCCESS in 12m 25s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/73/check/real-db-upgrade_nova_percona_user_002:th-percona/3206143507954338b6628faf8c2277da/index.html : SUCCESS in 13m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":73},{"id":"1efe31ceb6ff6cfc29beb225a7c9798ef29133ab","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-12 17:20:37.000000000","message":"Patch Set 73:\n\n* pci-test http://52.27.155.124/pci/193135/73 : SUCCESS","accounts_in_message":[],"_revision_number":73},{"id":"760895b0549255f9b77c62f0c6cbb16e679c5449","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-12 17:36:12.000000000","message":"Patch Set 73:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/73/check/check-ibm-tempest-dsvm-full/fe47922/ : SUCCESS in 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"3f5d44e25092a8add4e98064d6f87fa715edce41","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-12 17:42:26.000000000","message":"Patch Set 73: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/73/22796/index.html : SUCCESS in 59 m 46 s","accounts_in_message":[],"_revision_number":73},{"id":"d140fb3d210ee0b5dec34fe7da0c53cdefa1a61a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-12 17:55:17.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/73/check/dsvm-tempest-xen/2846b1a : FAILURE in 1h 16m 56s","accounts_in_message":[],"_revision_number":73},{"id":"63d2a0b6222b1fee7efc8235ebc7127ec0926728","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-12 18:07:38.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/73/check/gate-nova-pep8/e236616/ : SUCCESS in 9m 18s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/73/check/gate-nova-docs/76142dc//doc/build/html/ : SUCCESS in 14m 00s\n- gate-nova-python27 http://logs.openstack.org/35/193135/73/check/gate-nova-python27/02d722e/ : FAILURE in 10m 55s\n- gate-nova-python34 http://logs.openstack.org/35/193135/73/check/gate-nova-python34/3bf40f9/ : FAILURE in 32m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-full/ed117c8/ : SUCCESS in 57m 24s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-postgres-full/9557c8c/ : SUCCESS in 54m 55s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-neutron-full/f7d8443/ : SUCCESS in 1h 07m 32s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/73/check/gate-grenade-dsvm/ef07f8c/ : SUCCESS in 41m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-large-ops/d9366bd/ : SUCCESS in 30m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-neutron-large-ops/25171d0/ : SUCCESS in 31m 38s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/73/check/gate-nova-releasenotes/eadbfa7//releasenotes/build/html/ : SUCCESS in 8m 57s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/73/check/gate-nova-tox-functional/0fb5784/ : SUCCESS in 16m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/73/check/gate-grenade-dsvm-multinode/41495e1/ : SUCCESS in 49m 00s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/4f74724/ : FAILURE in 41m 39s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-nova-v20-api/81095c4/ : SUCCESS in 36m 43s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-nova-v20-api-legacy/22f6748/ : SUCCESS in 34m 39s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-cells/202c11a/ : SUCCESS in 39m 45s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-full-ceph/b8bc1be/ : SUCCESS in 55m 33s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/e2ceff3/ : SUCCESS in 1h 00m 28s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/73/check/gate-tempest-dsvm-multinode-full/10b3de2/ : SUCCESS in 55m 33s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/73/check/gate-compute-api-guide/51877f3//api-guide/build/html/ : SUCCESS in 9m 35s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/73/check/gate-nova-commit-message/b64ff4d/ : SUCCESS in 1m 50s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/73/check/gate-nova-python27-constraints/0280abe/ : SUCCESS in 14m 14s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/73/check/gate-nova-python34-constraints/5f4f3ee/ : SUCCESS in 28m 24s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/73/check/gate-nova-docs-constraints/989d41f//doc/build/html/ : SUCCESS in 15m 47s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/73/check/gate-nova-pep8-constraints/e99958e/ : SUCCESS in 9m 28s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/73/check/gate-nova-tox-constraints-functional-constraints/dcc9a50/ : SUCCESS in 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"a908446e0d8270653546b5f93304e4d6630b9782","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-12 20:19:45.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"8ec08b2c47724b0f0ddfc0a312e9bfdee8821ab3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-12 20:25:28.000000000","message":"Patch Set 74:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/74/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":74},{"id":"18cff48f903f0a2e929c6c3b8f3eaa75b734c8aa","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-12 20:54:55.000000000","message":"Patch Set 74: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"c25213ac0e8412d416bcb9f904aa83fc1975b2d2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-12 21:16:30.000000000","message":"Patch Set 74: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"18efd6d3120c6a01f1679fb23ef43d357cf021a9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-12 21:16:56.000000000","message":"Patch Set 74:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/74/check/check-ibm-tempest-dsvm-full/ad322e6/ : SUCCESS in 55m 49s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"09231aee713753fdba2fd38be523e654452ef744","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-12 21:17:15.000000000","message":"Patch Set 74: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"d7c7942391f4c5c042eb1eaf66e6d376e9d22211","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-12 21:22:31.000000000","message":"Patch Set 74:\n\n(2 comments)","accounts_in_message":[],"_revision_number":74},{"id":"7c38140cf400daa5a15139daa632e0d26831b588","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-12 21:26:00.000000000","message":"Patch Set 74: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/74/22850/index.html : SUCCESS in 56 m 37 s","accounts_in_message":[],"_revision_number":74},{"id":"56bb24567a90e7bd070da16da904c49dbda55c95","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-12 21:44:29.000000000","message":"Patch Set 74: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/74/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b8014b58a952451ea008a13e3f43706d/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/74/check/real-db-upgrade_nova_mysql_user_001:th-mysql/efea966d6de84c9abcd6053699cf6c28/index.html : SUCCESS in 16m 48s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/74/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c5cfcba7615e498ab602fdd844406577/index.html : SUCCESS in 15m 29s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/74/check/real-db-upgrade_nova_percona_user_001:th-percona/69446c401fde46079ef75a437c55012a/index.html : SUCCESS in 16m 45s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/74/check/real-db-upgrade_nova_mysql_user_002:th-mysql/17ab78e178b64ddfa44c0a34ffebf59f/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/74/check/real-db-upgrade_nova_percona_user_002:th-percona/eb50ae1ef74a415f88180c4b151a1383/index.html : SUCCESS in 14m 59s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":74},{"id":"5a28ef5fe7bf84674472fc59761c26e6e22ca558","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-12 21:47:47.000000000","message":"Patch Set 74:\n\n* pci-test http://52.27.155.124/pci/193135/74 : SUCCESS","accounts_in_message":[],"_revision_number":74},{"id":"dce769821aa95807f60d9136e29452484387ef39","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-12 23:02:20.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/74/check/dsvm-tempest-xen/d34de7e : SUCCESS in 1h 21m 31s","accounts_in_message":[],"_revision_number":74},{"id":"a8c1066de9ba8901d35c94eb623789fc93cc4d8a","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-12 23:16:23.000000000","message":"Patch Set 74:\n\n(4 comments)","accounts_in_message":[],"_revision_number":74},{"id":"4067eb4a1df97edd4ff52eb34a51f639af9005c6","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-12 23:39:39.000000000","message":"Uploaded patch set 75.","accounts_in_message":[],"_revision_number":75},{"id":"72a5a8a6695b865889fa689b0043d428ecd6df2a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-12 23:46:11.000000000","message":"Patch Set 75:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/75/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":75},{"id":"35448f6205d1fe28015f5c49826341078276bda4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-12 23:57:20.000000000","message":"Patch Set 75:\n\n* pci-test http://52.27.155.124/pci/193135/75 : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"2455f964101bfc204269ea10f6b04eb47581a6cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-13 00:42:46.000000000","message":"Patch Set 75:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/75/check/check-ibm-tempest-dsvm-full/13fcc00/ : FAILURE in 1h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"e5f39fd788c8527202593b48134e2bb2a4bc302c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-13 00:44:21.000000000","message":"Patch Set 75: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/75/22878/index.html : FAILURE in 01 h 02 m 14 s","accounts_in_message":[],"_revision_number":75},{"id":"e7b2fab2709a2eb5ff907d6e2fc4b430f390b762","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-13 00:48:29.000000000","message":"Patch Set 75: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/75/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6916843ccad145a89789f14e930b1ec8/index.html : SUCCESS in 13m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/75/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0d3eba9141c747b592fb4760b78cb9c6/index.html : SUCCESS in 15m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/75/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f80405b4ad5442b384337cce60b84a80/index.html : SUCCESS in 17m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/75/check/real-db-upgrade_nova_percona_user_001:th-percona/9273d58ff0e845df981ff2b7459edb8f/index.html : SUCCESS in 15m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/75/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a74e97bb1acd4434b7701140ad13ca1c/index.html : SUCCESS in 13m 10s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/75/check/real-db-upgrade_nova_percona_user_002:th-percona/8536e0f5ea5a4a7d8d431607f0cba936/index.html : SUCCESS in 12m 59s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":75},{"id":"9d292531f709b61da407f72babd3647846f641a5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-13 01:05:24.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/75/check/gate-nova-pep8/dfa7bba/ : SUCCESS in 7m 10s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/75/check/gate-nova-docs/8ee5727//doc/build/html/ : SUCCESS in 15m 48s\n- gate-nova-python27 http://logs.openstack.org/35/193135/75/check/gate-nova-python27/d32cd64/ : SUCCESS in 13m 43s\n- gate-nova-python34 http://logs.openstack.org/35/193135/75/check/gate-nova-python34/3df194c/ : FAILURE in 28m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-full/382e6b7/ : FAILURE in 47m 00s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-postgres-full/b43edc2/ : FAILURE in 51m 12s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-neutron-full/03ba0f6/ : FAILURE in 1h 15m 11s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/75/check/gate-grenade-dsvm/d3c77f1/ : FAILURE in 47m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-large-ops/2594629/ : SUCCESS in 42m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-neutron-large-ops/804f40f/ : SUCCESS in 36m 51s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/75/check/gate-nova-releasenotes/e01841f//releasenotes/build/html/ : SUCCESS in 7m 21s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/75/check/gate-nova-tox-functional/a0b91e7/ : SUCCESS in 12m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/75/check/gate-grenade-dsvm-multinode/1a184ce/ : FAILURE in 1h 00m 46s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b429e60/ : SUCCESS in 48m 00s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-nova-v20-api/6d7364d/ : FAILURE in 43m 03s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-nova-v20-api-legacy/853cfce/ : FAILURE in 43m 25s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-cells/b157e47/ : FAILURE in 50m 39s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-full-ceph/97ef620/ : FAILURE in 58m 07s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4aca06c/ : FAILURE in 1h 01m 42s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-multinode-full/4cec3b6/ : FAILURE in 1h 01m 10s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/75/check/gate-compute-api-guide/e65af0d//api-guide/build/html/ : SUCCESS in 9m 25s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/75/check/gate-nova-commit-message/2f08abc/ : SUCCESS in 1m 28s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/75/check/gate-nova-python27-constraints/9f2c413/ : SUCCESS in 11m 44s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/75/check/gate-nova-python34-constraints/490a0b5/ : SUCCESS in 29m 32s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/75/check/gate-nova-docs-constraints/c266133//doc/build/html/ : SUCCESS in 15m 16s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/75/check/gate-nova-pep8-constraints/b8d9be8/ : SUCCESS in 9m 49s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/75/check/gate-nova-tox-constraints-functional-constraints/a648756/ : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"2ff8ebcc9cc31f4495d8d9eb7d9e85be2f7e82f8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-13 02:05:30.000000000","message":"Patch Set 75: Code-Review-1\n\n(1 comment)\n\nI haven\u0027t fully reviewed this yet, still working on the testing changes that start with devstack here: https://review.openstack.org/#/c/266633/","accounts_in_message":[],"_revision_number":75},{"id":"c89d22fcaf62aea4f4f73dd6ea12c5b88bdb4028","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-13 02:06:36.000000000","message":"Patch Set 75:\n\nThere is a tempest test failure that is related here:\n\nhttp://logs.openstack.org/35/193135/75/check/gate-tempest-dsvm-full/382e6b7/console.html#_2016-01-13_00_36_17_242\n\nSo tempest is probably going to need to be updated for this also (as a dependency).","accounts_in_message":[],"_revision_number":75},{"id":"0853e718f5f6922cf193a3044bcf0a9d33e98593","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-13 02:17:32.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/75/check/dsvm-tempest-xen/60da1db : SUCCESS in 2h 36m 49s","accounts_in_message":[],"_revision_number":75},{"id":"566884d19dfdc0a2e8013536589d7016b9565f40","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 03:11:57.000000000","message":"Patch Set 75:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/75/ : FAILURE in 3h 26m 55s","accounts_in_message":[],"_revision_number":75},{"id":"ab2286658b8c37006f58f07681ff9e63afeb8c54","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-13 05:02:41.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/75/Nova-ML2-Sriov : SUCCESS in 28m 03s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":75},{"id":"0de8ee0527863fc20f605119639120b4d3a9b14d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 05:17:49.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/75/check/check-dsvm-tempest-vz7-exe-minimal/ba6a0d8 : SUCCESS in 52m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":75},{"id":"39fdda9cbe7406a572e7a21157ccd15a7f1b1739","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-13 06:51:10.000000000","message":"Uploaded patch set 76: Patch Set 75 was rebased.","accounts_in_message":[],"_revision_number":76},{"id":"ccee90f35d05fd9e94247d6bcd6153aeb501cf5f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-13 07:19:21.000000000","message":"Patch Set 76: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/76/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/55b3cd6a86ab44c5a595839ff64f9f4b/index.html : SUCCESS in 12m 53s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/76/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e961f27b5e7a4ea3baef1df27f3f65d9/index.html : SUCCESS in 16m 44s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/76/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a8088dcc89694b9cb9bcfe327d854502/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/76/check/real-db-upgrade_nova_percona_user_001:th-percona/faf53033b9694a9ebebdd1d249e02cf8/index.html : SUCCESS in 16m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/76/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6239a1bab5aa4713a927c0d137165274/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/76/check/real-db-upgrade_nova_percona_user_002:th-percona/0c4f5af76f43429ca5c1f0cc28b9dc75/index.html : SUCCESS in 14m 46s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":76},{"id":"aa52766c9545804a3cfb961658a7c19c7316292a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-13 07:28:28.000000000","message":"Patch Set 76:\n\n* pci-test http://52.27.155.124/pci/193135/76 : SUCCESS","accounts_in_message":[],"_revision_number":76},{"id":"165bed8ab1ed9e716b5ac7c7d621fd70986448fc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-13 07:53:54.000000000","message":"Patch Set 76:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/76/check/check-ibm-tempest-dsvm-full/7dba132/ : FAILURE in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"2512d259c7ad4785f7863b64239567b65f1d0d9d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-13 08:11:27.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/76/check/gate-nova-pep8/3b0f327/ : SUCCESS in 7m 48s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/76/check/gate-nova-docs/a3f9f0f//doc/build/html/ : SUCCESS in 15m 46s\n- gate-nova-python27 http://logs.openstack.org/35/193135/76/check/gate-nova-python27/21dc02b/ : SUCCESS in 12m 25s\n- gate-nova-python34 http://logs.openstack.org/35/193135/76/check/gate-nova-python34/c1711c5/ : SUCCESS in 27m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-full/696c935/ : FAILURE in 56m 21s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-postgres-full/b516c01/ : FAILURE in 1h 03m 41s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-neutron-full/019a461/ : FAILURE in 1h 05m 11s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/76/check/gate-grenade-dsvm/b967897/ : FAILURE in 52m 01s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-large-ops/3baf10f/ : SUCCESS in 37m 36s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-neutron-large-ops/5eefcc3/ : SUCCESS in 34m 37s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/76/check/gate-nova-releasenotes/6ca0a23//releasenotes/build/html/ : SUCCESS in 7m 26s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/76/check/gate-nova-tox-functional/68b747f/ : SUCCESS in 14m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/76/check/gate-grenade-dsvm-multinode/cfb9554/ : FAILURE in 55m 02s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6dfbabf/ : FAILURE in 27m 38s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-nova-v20-api/9900d34/ : FAILURE in 35m 43s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-nova-v20-api-legacy/15eb6d1/ : FAILURE in 36m 33s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-cells/d58f33e/ : FAILURE in 49m 51s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-full-ceph/9023943/ : FAILURE in 1h 08m 54s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/a45ddc0/ : FAILURE in 1h 13m 16s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/76/check/gate-tempest-dsvm-multinode-full/e03a855/ : FAILURE in 1h 08m 40s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/76/check/gate-compute-api-guide/34ab297//api-guide/build/html/ : SUCCESS in 8m 35s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/76/check/gate-nova-commit-message/65532a4/ : SUCCESS in 1m 54s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/76/check/gate-nova-python27-constraints/a77ae67/ : SUCCESS in 13m 08s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/76/check/gate-nova-python34-constraints/5a71fe4/ : SUCCESS in 26m 12s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/76/check/gate-nova-docs-constraints/62e777c//doc/build/html/ : SUCCESS in 11m 27s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/76/check/gate-nova-pep8-constraints/dd36a0b/ : SUCCESS in 8m 06s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/76/check/gate-nova-tox-constraints-functional-constraints/6395565/ : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"05e9ca9779f2db3477e4da100a695612f84e0edf","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-13 08:12:50.000000000","message":"Patch Set 76: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/76/22936/index.html : FAILURE in 01 h 09 m 44 s","accounts_in_message":[],"_revision_number":76},{"id":"309876a34695b4cff77828805f07ff44144c6875","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-13 08:16:05.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/76/check/dsvm-tempest-xen/37b7a2a : SUCCESS in 1h 23m 35s","accounts_in_message":[],"_revision_number":76},{"id":"2e5303d99d2aaa549e8ff540ed609bac5057aae6","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-13 09:21:18.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"51f319c04d3e899e295e514df7026a976319a515","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-13 09:39:46.000000000","message":"Patch Set 77:\n\n* pci-test http://52.27.155.124/pci/193135/77 : SUCCESS","accounts_in_message":[],"_revision_number":77},{"id":"d83598c011c2ae47346cda349f4adbde911d7e1a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-13 09:42:46.000000000","message":"Patch Set 77: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/77/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d84062f9660b4ab79312a8cac8d6ff13/index.html : SUCCESS in 16m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/77/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d459cfce2aa443229c9d497b2dcf5eb6/index.html : SUCCESS in 16m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/77/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/77acb7a7e9984dceacf4c1d7868291a3/index.html : SUCCESS in 15m 29s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/77/check/real-db-upgrade_nova_percona_user_001:th-percona/9355ae0f4fd34331bbe5392dedcb04c5/index.html : SUCCESS in 17m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/77/check/real-db-upgrade_nova_mysql_user_002:th-mysql/71ea26a408044847bdd12fbd3a17a2f0/index.html : SUCCESS in 12m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/77/check/real-db-upgrade_nova_percona_user_002:th-percona/25c2aeed8b224111b7b90aee3b676a13/index.html : SUCCESS in 14m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":77},{"id":"0870d88613d07ece1a7184de45d5e4dfcfb5cd58","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-13 10:16:22.000000000","message":"Patch Set 77:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/77/check/check-ibm-tempest-dsvm-full/1453b27/ : FAILURE in 53m 54s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"3cf958ed65516cbf5b5eb465de1b21b3ca46441c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-13 10:27:37.000000000","message":"Patch Set 77: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/77/22955/index.html : SUCCESS in 01 h 00 m 43 s","accounts_in_message":[],"_revision_number":77},{"id":"0efb32f1603d2dcada5aa387c56539e2594ed66b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-13 10:27:52.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/77/check/dsvm-tempest-xen/2782cd0 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":77},{"id":"9df6b2015759ba1a2b813d0a135b2994ca0295bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-13 10:37:18.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/77/check/gate-nova-pep8/46f2f34/ : SUCCESS in 10m 58s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/77/check/gate-nova-docs/382c2db//doc/build/html/ : SUCCESS in 11m 36s\n- gate-nova-python27 http://logs.openstack.org/35/193135/77/check/gate-nova-python27/4583f0d/ : SUCCESS in 13m 57s\n- gate-nova-python34 http://logs.openstack.org/35/193135/77/check/gate-nova-python34/78081ac/ : SUCCESS in 27m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-full/837ab27/ : SUCCESS in 45m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-postgres-full/67f948c/ : SUCCESS in 49m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-neutron-full/48bb0b3/ : SUCCESS in 58m 30s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/77/check/gate-grenade-dsvm/abfc648/ : SUCCESS in 42m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-large-ops/4751b72/ : SUCCESS in 28m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-neutron-large-ops/514cc7e/ : SUCCESS in 33m 25s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/77/check/gate-nova-releasenotes/ae4dfaf//releasenotes/build/html/ : SUCCESS in 7m 39s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/77/check/gate-nova-tox-functional/c4dd744/ : SUCCESS in 15m 33s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/77/check/gate-grenade-dsvm-multinode/19eb212/ : FAILURE in 1h 07m 18s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/08f8677/ : SUCCESS in 44m 03s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-nova-v20-api/9441a14/ : SUCCESS in 46m 06s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-nova-v20-api-legacy/a4bddd6/ : SUCCESS in 34m 52s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-cells/b269902/ : SUCCESS in 36m 27s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-full-ceph/6230a7f/ : SUCCESS in 41m 02s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/ee461b6/ : SUCCESS in 51m 33s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/77/check/gate-tempest-dsvm-multinode-full/9294884/ : FAILURE in 1h 00m 24s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/77/check/gate-compute-api-guide/f32d849//api-guide/build/html/ : SUCCESS in 7m 03s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/77/check/gate-nova-commit-message/417c702/ : SUCCESS in 1m 23s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/77/check/gate-nova-python27-constraints/26519b3/ : SUCCESS in 15m 18s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/77/check/gate-nova-python34-constraints/d1abec1/ : SUCCESS in 28m 47s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/77/check/gate-nova-docs-constraints/2de54d4//doc/build/html/ : SUCCESS in 14m 16s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/77/check/gate-nova-pep8-constraints/530e97f/ : SUCCESS in 7m 39s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/77/check/gate-nova-tox-constraints-functional-constraints/d21da61/ : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"97f5948dc935c6f548a4ba9e25312c4baa392233","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-13 10:44:23.000000000","message":"Patch Set 77:\n\nrecheck","accounts_in_message":[],"_revision_number":77},{"id":"97eb59052d830937c134ff6121a8ad670ffce6db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-13 10:45:13.000000000","message":"Patch Set 77:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":77},{"id":"914719e469649cddaa101836fe340b50ebcf4c35","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 11:00:35.000000000","message":"Patch Set 77:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/77/ : FAILURE in 1h 37m 42s","accounts_in_message":[],"_revision_number":77},{"id":"da99b58df53cf1b12bf509410beb8b9b114b69db","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-13 11:49:54.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"ebdbc4aef76d0d5bafdaa0b6a907d2f00876b63b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-13 12:14:34.000000000","message":"Patch Set 78: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/78/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/968f5fa6b7db4de99a22a304602387c8/index.html : SUCCESS in 15m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/78/check/real-db-upgrade_nova_mysql_user_001:th-mysql/293ec530dca54bfda3bdd8a15efeb846/index.html : SUCCESS in 16m 05s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/78/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/41836483862648289c2ecfa85e42a10f/index.html : SUCCESS in 12m 44s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/78/check/real-db-upgrade_nova_percona_user_001:th-percona/e8ae747e996a4c0dadfd0a84faf7fc70/index.html : SUCCESS in 13m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/78/check/real-db-upgrade_nova_mysql_user_002:th-mysql/8ffb716cdee0472baea45b2265139dc8/index.html : SUCCESS in 15m 17s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/78/check/real-db-upgrade_nova_percona_user_002:th-percona/d4a077a89ab248988d9261d111f1f320/index.html : SUCCESS in 15m 35s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":78},{"id":"aee595cf97e249e94e20e1d16c1d06cfecb3e009","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-13 12:17:39.000000000","message":"Patch Set 78:\n\n* pci-test http://52.27.155.124/pci/193135/78 : SUCCESS","accounts_in_message":[],"_revision_number":78},{"id":"c3cd8747872949ada6c25c4a9a05ebd76e464e7d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-13 12:47:53.000000000","message":"Patch Set 78:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/78/check/check-ibm-tempest-dsvm-full/71adb95/ : FAILURE in 56m 52s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"adb4ecb25924647605192601a47812c547ec6128","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-13 12:54:04.000000000","message":"Patch Set 78: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/78/22978/index.html : SUCCESS in 01 h 00 m 49 s","accounts_in_message":[],"_revision_number":78},{"id":"56d97c0c31719bc8f8b4c839420b23b8da995e75","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-13 12:57:36.000000000","message":"Patch Set 75:\n\n(1 comment)\n\nHi Matt,\n\nThanks for the review. I added docs in the next patch set, also fixed the failure, it happened due to a messed up function call.\n\nBest Regards,\nIldiko","accounts_in_message":[],"_revision_number":75},{"id":"e5a1dc21456de4da2f1b6d50da7d4798a4b54855","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-13 12:58:08.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/78/check/dsvm-tempest-xen/1b92c02 : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":78},{"id":"2a24e982aae0d8f298202dc945801b5e5c778b5e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 12:58:40.000000000","message":"Patch Set 78:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/78/ : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":78},{"id":"fccb982620347b5c39714ec098f8e473d549663b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-13 13:26:29.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/78/check/gate-nova-pep8/46acc0a/ : SUCCESS in 9m 22s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/78/check/gate-nova-docs/2fe46aa//doc/build/html/ : SUCCESS in 14m 44s\n- gate-nova-python27 http://logs.openstack.org/35/193135/78/check/gate-nova-python27/d6e1aba/ : SUCCESS in 15m 53s\n- gate-nova-python34 http://logs.openstack.org/35/193135/78/check/gate-nova-python34/311fc6f/ : SUCCESS in 24m 52s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-full/420133f/ : SUCCESS in 39m 57s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-postgres-full/2408ba8/ : SUCCESS in 44m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-neutron-full/b0d83eb/ : SUCCESS in 1h 31m 46s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/78/check/gate-grenade-dsvm/922e590/ : SUCCESS in 56m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-large-ops/34dd940/ : SUCCESS in 41m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-neutron-large-ops/834b1f7/ : SUCCESS in 29m 21s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/78/check/gate-nova-releasenotes/1e07d8f//releasenotes/build/html/ : SUCCESS in 9m 49s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/78/check/gate-nova-tox-functional/a2fdd6a/ : SUCCESS in 17m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/78/check/gate-grenade-dsvm-multinode/85951d6/ : SUCCESS in 51m 18s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/24f4d56/ : FAILURE in 35m 38s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-nova-v20-api/2857bdc/ : SUCCESS in 30m 34s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-nova-v20-api-legacy/62af6cf/ : SUCCESS in 42m 06s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-cells/c81b0b0/ : SUCCESS in 34m 34s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-full-ceph/c293f3b/ : SUCCESS in 54m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/2d1193c/ : SUCCESS in 1h 20m 48s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/78/check/gate-tempest-dsvm-multinode-full/1562b68/ : SUCCESS in 57m 05s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/78/check/gate-compute-api-guide/f09a8c1//api-guide/build/html/ : SUCCESS in 8m 53s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/35/193135/78/check/gate-nova-commit-message/ede08d8/ : SUCCESS in 2m 42s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/78/check/gate-nova-python27-constraints/118e6f9/ : SUCCESS in 14m 13s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/78/check/gate-nova-python34-constraints/9ad7cea/ : SUCCESS in 28m 26s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/78/check/gate-nova-docs-constraints/e198e0c//doc/build/html/ : SUCCESS in 14m 46s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/78/check/gate-nova-pep8-constraints/10d5a9b/ : SUCCESS in 9m 48s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/78/check/gate-nova-tox-constraints-functional-constraints/a5deab8/ : SUCCESS in 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"def7fce8992975f43651aa98da61aeeb05f1eb83","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 13:51:13.000000000","message":"Patch Set 77:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/77/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":77},{"id":"4aa0a0a759edfaaadfdbebd6efa621b948d967e6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 16:13:53.000000000","message":"Patch Set 78:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/78/check/check-dsvm-tempest-vz7-exe-minimal/d4a305d : FAILURE in 1h 24m 56s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":78},{"id":"7938c9ce9b3c0c7eb263591da1d35ed22a9968b1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 19:09:07.000000000","message":"Patch Set 77:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/77/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":77},{"id":"906b48a2e7997cb4cd3c36919f0b4567f1629c0c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-13 23:21:52.000000000","message":"Patch Set 78:\n\nBuild Failed.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/78/compute-nfv-flavors : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":78},{"id":"6ff02cb841d0f660826a59c8841e08094349ccd7","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-14 14:08:47.000000000","message":"Uploaded patch set 79: Patch Set 78 was rebased.","accounts_in_message":[],"_revision_number":79},{"id":"cd5c2a75eb2105789500cfd09b93420ea77b1f75","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-14 14:40:11.000000000","message":"Patch Set 79:\n\n* pci-test http://52.27.155.124/pci/193135/79 : SUCCESS","accounts_in_message":[],"_revision_number":79},{"id":"56d97a72b44372d5922786c77b92b48e9c98feca","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-14 14:40:23.000000000","message":"Patch Set 79: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":79},{"id":"e43cd1f9aa1a9ce70caba2e34e9cbf0411c889ee","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-14 14:40:24.000000000","message":"Patch Set 79: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":79},{"id":"44d8696396dc508afe785be9d280c5d42e77e2c0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-14 14:40:24.000000000","message":"Patch Set 79:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":79},{"id":"3230f87a579aaba03674376fa84fb4aed906257b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-14 14:40:25.000000000","message":"Patch Set 79:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":79},{"id":"b095a25522f998ee011319ba3d6b6616e1692534","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-14 14:40:26.000000000","message":"Patch Set 79:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":79},{"id":"9d72a94c4cf622ba282ab1ad42c5bded7ea638b6","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-14 15:10:00.000000000","message":"Patch Set 79: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/79/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6c4d510452d44cffbb8ead281b102772/index.html : SUCCESS in 13m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/79/check/real-db-upgrade_nova_mysql_user_001:th-mysql/40c6966bed374947a0de32f6776be706/index.html : SUCCESS in 15m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/79/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/520e529ced4d46bf8a0012449802d127/index.html : SUCCESS in 12m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/79/check/real-db-upgrade_nova_percona_user_001:th-percona/82618121b1b44abba12dc5a757e3f352/index.html : SUCCESS in 18m 44s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/79/check/real-db-upgrade_nova_mysql_user_002:th-mysql/42c54911488a4346a8746889189725ab/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/79/check/real-db-upgrade_nova_percona_user_002:th-percona/371fd500901f47d29d64245fd89fde79/index.html : SUCCESS in 15m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":79},{"id":"8c2bd5035cdc0f6ab44a113172f2a13b313c25e6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-14 15:43:01.000000000","message":"Patch Set 79: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/79/2796/index.html : SUCCESS in 59 m 05 s","accounts_in_message":[],"_revision_number":79},{"id":"c0ec8cec4e2a56205a5972914eb162ed8a67eef4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-14 16:26:11.000000000","message":"Patch Set 79:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/79/check/check-dsvm-tempest-vz7-exe-minimal/856b88a : SUCCESS in 1h 28m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":79},{"id":"e6bd88b56ef9ff5362b68009fff57a7e1eae9621","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-15 13:29:15.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"3e1faf2fa6b97ca8d4817b5fe848cd91d5a5bf70","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-15 13:29:39.000000000","message":"Patch Set 80: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":80},{"id":"ccf07f1b0906c04bd24957d05bd2825ed102f7c4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-15 13:29:47.000000000","message":"Patch Set 80:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":80},{"id":"405c49a31b35aefd70209d7c2ed7018894b0ba0f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-15 13:29:48.000000000","message":"Patch Set 80:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":80},{"id":"9f5106d85f307bf0b363cce1f48f4dc03f768762","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-15 13:30:31.000000000","message":"Patch Set 80:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":80},{"id":"0f4aba30c6651eb3ee25005193fea5a24e6dcde6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-15 13:46:31.000000000","message":"Patch Set 80:\n\n* pci-test http://52.27.155.124/pci/193135/80 : SUCCESS","accounts_in_message":[],"_revision_number":80},{"id":"2e912984bff1b3e7c9af5245987c9574ddcca56a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-15 13:56:30.000000000","message":"Patch Set 80: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/80/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bd49f149e4424f399e508173a88e0b03/index.html : SUCCESS in 15m 55s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/80/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ba28c6345f8f4e44844717e45b97ec80/index.html : SUCCESS in 17m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/80/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3fa5682874034e9483de60c72843670f/index.html : SUCCESS in 12m 31s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/80/check/real-db-upgrade_nova_percona_user_001:th-percona/81e658800981425d9621c55dc252bf32/index.html : SUCCESS in 17m 08s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/80/check/real-db-upgrade_nova_mysql_user_002:th-mysql/eaef61059b2f4bfa8afe9e5f3dcf52de/index.html : SUCCESS in 15m 09s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/80/check/real-db-upgrade_nova_percona_user_002:th-percona/ee3736dc7c2a4637940e7f788d3c7c74/index.html : SUCCESS in 14m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":80},{"id":"9d2899ef4d533db9d5edc3d88017b593912174e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-15 13:58:41.000000000","message":"Patch Set 80: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":80},{"id":"59e37e5002c648acb52a93365ffe774dff050798","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-15 14:37:58.000000000","message":"Patch Set 80:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/80/check/check-ibm-tempest-dsvm-full/2647406/ : SUCCESS in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"ce06b4a93e9eafcba6b959555b985339df073688","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-15 14:45:56.000000000","message":"Patch Set 80: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/80/3018/index.html : SUCCESS in 01 h 11 m 05 s","accounts_in_message":[],"_revision_number":80},{"id":"6f5e2fe931a173ebc26abfbe27f99ab324fe0979","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-15 22:44:17.000000000","message":"Uploaded patch set 81: Patch Set 80 was rebased.","accounts_in_message":[],"_revision_number":81},{"id":"d2f5b6dd41f47b251f3d68263831f0e3a0420183","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-15 22:44:37.000000000","message":"Patch Set 81: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":81},{"id":"f0cb5824ed5aa92ebc566f13b04e949f858cae36","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-15 22:44:38.000000000","message":"Patch Set 81: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":81},{"id":"f9d43f7e5d7f7e06dede1cea4c6de2b2545a7994","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-15 22:44:41.000000000","message":"Patch Set 81:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":81},{"id":"48b3258bb1f26f1d50068a93820d8d1f3e4fea2d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-15 22:44:45.000000000","message":"Patch Set 81:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":81},{"id":"bcea7460bf0c40dc6078b27f1b77669b8948813f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-15 22:44:47.000000000","message":"Patch Set 81:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":81},{"id":"2b1b396e8286588450fa4e158051b43f9bead758","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-15 22:46:06.000000000","message":"Patch Set 81:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":81},{"id":"1935195e4f8eff216732aae16e4a3f9828af1a8d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-15 23:03:02.000000000","message":"Patch Set 81:\n\n* pci-test http://52.27.155.124/pci/193135/81 : SUCCESS","accounts_in_message":[],"_revision_number":81},{"id":"f2766eb9a96c958d97113c3f7d281d57feb178bd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-15 23:10:25.000000000","message":"Patch Set 81: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/81/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/86daa0962ef64a5783bbd72fefb1aa45/index.html : SUCCESS in 14m 07s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/81/check/real-db-upgrade_nova_mysql_user_001:th-mysql/55d03eec738a40408538d9472d0605cd/index.html : SUCCESS in 17m 01s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/81/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d9ccfdceedd34280916201c45857d675/index.html : SUCCESS in 17m 07s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/81/check/real-db-upgrade_nova_percona_user_001:th-percona/3a9d9ff966034f2390f0851cf39de298/index.html : SUCCESS in 16m 00s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/81/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c4d44c7d6fc643f4832ccd364662797c/index.html : SUCCESS in 13m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/81/check/real-db-upgrade_nova_percona_user_002:th-percona/5800284581744fada678f37565415a6f/index.html : SUCCESS in 15m 14s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":81},{"id":"068d2325c726f4bcdb395fcdeba66668e15c63de","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-16 00:03:49.000000000","message":"Patch Set 81: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/81/3083/index.html : SUCCESS in 01 h 13 m 58 s","accounts_in_message":[],"_revision_number":81},{"id":"15ce212cfbc914f512e5c1b130bb675756237dff","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-16 10:09:18.000000000","message":"Uploaded patch set 82: Patch Set 81 was rebased.","accounts_in_message":[],"_revision_number":82},{"id":"d142632a80ecbb5e3a589afe08e65ca3239ff7f1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-16 10:09:40.000000000","message":"Patch Set 82: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":82},{"id":"82e448cad0d246ad318adcb34fba51c9e6ada684","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-16 10:09:43.000000000","message":"Patch Set 82:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":82},{"id":"ac3c5b4176222177849109e55d6e1b883257223c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-16 10:09:51.000000000","message":"Patch Set 82:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":82},{"id":"e3e87a854ee851aae658f74211e08d330c60d958","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 10:09:59.000000000","message":"Patch Set 82:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":82},{"id":"25fbe024121a3f67f3cb74e8abb234bb4d7ad17e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-16 10:25:04.000000000","message":"Patch Set 82:\n\n* pci-test http://52.27.155.124/pci/193135/82 : SUCCESS","accounts_in_message":[],"_revision_number":82},{"id":"0135936624d86c30e2eede3c17ae07b5cd267ec2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-16 10:26:47.000000000","message":"Patch Set 82: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":82},{"id":"6029e252a5d778b5977549150469d6f7cf22c1d1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-16 10:28:48.000000000","message":"Patch Set 82: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/82/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5f79b786c6d24ff7a9a1450ac9c14dfc/index.html : SUCCESS in 12m 32s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/82/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a567dd402d274bcb91b6891e76727776/index.html : SUCCESS in 15m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/82/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/60313b1f7d3140cbab176ff19c3ff08e/index.html : SUCCESS in 14m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/82/check/real-db-upgrade_nova_percona_user_001:th-percona/b4984e270bbc4bb88d05495e231ff0e6/index.html : SUCCESS in 16m 38s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/82/check/real-db-upgrade_nova_mysql_user_002:th-mysql/dbaa72d3d6104dccb07cf542031f5744/index.html : SUCCESS in 12m 55s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/82/check/real-db-upgrade_nova_percona_user_002:th-percona/74fe114634ac42369859ceb5948fcc56/index.html : SUCCESS in 14m 31s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":82},{"id":"5b5ba77cc9252fbfdab26589faf8509012a2d6ed","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 11:05:06.000000000","message":"Patch Set 82:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/82/check/check-dsvm-tempest-vz7-exe-minimal/7ba617c : FAILURE in 53m 15s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":82},{"id":"dd0163f28cc09cdaa850164ae03234ca91038529","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-16 11:22:39.000000000","message":"Patch Set 82: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/82/3116/index.html : SUCCESS in 01 h 09 m 04 s","accounts_in_message":[],"_revision_number":82},{"id":"2069299bba2a2db3c1b6e6d5d3356a4e934332c2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 12:31:28.000000000","message":"Patch Set 80:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/80/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":80},{"id":"778ea8376f1f1f42eb8b0c9b8a269fa759168074","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-16 17:12:49.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"5671c133ca1b75e64a9572cec290d8403e6d4c42","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-16 17:28:54.000000000","message":"Patch Set 83:\n\n* pci-test http://52.27.155.124/pci/193135/83 : SUCCESS","accounts_in_message":[],"_revision_number":83},{"id":"2126f6907663f8e717a5e837384cbdf5e56b62eb","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-16 17:35:18.000000000","message":"Patch Set 83: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/83/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/771645bd505f4649b3c9b5f8ec938feb/index.html : SUCCESS in 12m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/83/check/real-db-upgrade_nova_mysql_user_001:th-mysql/73d1f7f2e2cc48428d5001bd6cdc9cae/index.html : SUCCESS in 17m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/83/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c42b32e7837747dfb516953315af65c6/index.html : SUCCESS in 15m 25s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/83/check/real-db-upgrade_nova_percona_user_001:th-percona/2165da7e842440cb9da864c8c99752b6/index.html : SUCCESS in 17m 25s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/83/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4c391ae5af9143f4ae290eecce6a0af4/index.html : SUCCESS in 15m 20s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/83/check/real-db-upgrade_nova_percona_user_002:th-percona/0e10d20573374a7a9c9ceb8d0d7a8179/index.html : SUCCESS in 13m 04s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":83},{"id":"4494a53addf838a7e7e6cfdf81043d7b9caa63ca","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 18:11:06.000000000","message":"Patch Set 83:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/83/check/check-dsvm-tempest-vz7-exe-minimal/49d6f89 : FAILURE in 56m 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":83},{"id":"554a28f3245bd84b27f27e5e2e3910719c84bb06","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-16 18:21:47.000000000","message":"Patch Set 83:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/83/check/check-ibm-tempest-dsvm-full/89dd560/ : SUCCESS in 1h 07m 27s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"fa92783920d8d3c6a7c6f0879933e1a860cb49e0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-16 18:24:43.000000000","message":"Patch Set 83: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/83/3142/index.html : SUCCESS in 01 h 08 m 11 s","accounts_in_message":[],"_revision_number":83},{"id":"767f4a9173d6bb8f8c96428aeaa8091a88f5b31b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-16 18:41:56.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/83/check/dsvm-tempest-xen/7934c49 : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":83},{"id":"b1e7fd06e2f43ed34e7aa9c969d97a25dbb633e9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 19:35:39.000000000","message":"Patch Set 83:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/83/ : FAILURE in 2h 20m 20s","accounts_in_message":[],"_revision_number":83},{"id":"aedd6d71df0db3ae3f1e15d764e5c8eaf18efeb5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-16 19:49:06.000000000","message":"Patch Set 83: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/83/check/gate-nova-pep8/20d512c/ : SUCCESS in 9m 45s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/83/check/gate-nova-docs/3ae2c58//doc/build/html/ : SUCCESS in 14m 02s\n- gate-nova-python27 http://logs.openstack.org/35/193135/83/check/gate-nova-python27/f954392/ : SUCCESS in 14m 58s\n- gate-nova-python34 http://logs.openstack.org/35/193135/83/check/gate-nova-python34/43fef89/ : SUCCESS in 28m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-full/06cc155/ : SUCCESS in 59m 47s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-postgres-full/b7216c8/ : SUCCESS in 50m 13s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-neutron-full/2078f48/ : SUCCESS in 1h 24m 10s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/83/check/gate-grenade-dsvm/1de20d8/ : FAILURE in 59m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-large-ops/e2ca3d5/ : SUCCESS in 34m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-neutron-large-ops/c6929f3/ : SUCCESS in 33m 43s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/83/check/gate-nova-releasenotes/5f08272//releasenotes/build/html/ : SUCCESS in 9m 44s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/83/check/gate-nova-tox-functional/574abc4/ : SUCCESS in 16m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/83/check/gate-grenade-dsvm-multinode/a6e9de6/ : SUCCESS in 53m 47s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/4997307/ : SUCCESS in 49m 29s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-nova-v20-api/8c8c839/ : SUCCESS in 51m 44s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-nova-v20-api-legacy/4176b7c/ : SUCCESS in 51m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-cells/fffbf6b/ : SUCCESS in 55m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-full-ceph/854012a/ : SUCCESS in 1h 01m 35s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5031049/ : SUCCESS in 1h 26m 34s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/83/check/gate-tempest-dsvm-multinode-full/53c9404/ : FAILURE in 1h 00m 57s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/83/check/gate-compute-api-guide/8e1f022//api-guide/build/html/ : SUCCESS in 9m 43s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/83/check/gate-nova-python27-constraints/c2cf88a/ : SUCCESS in 14m 17s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/83/check/gate-nova-python34-constraints/241a915/ : SUCCESS in 28m 39s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/83/check/gate-nova-docs-constraints/f010b3c//doc/build/html/ : SUCCESS in 14m 41s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/83/check/gate-nova-pep8-constraints/65046a9/ : SUCCESS in 9m 47s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/83/check/gate-nova-tox-constraints-functional-constraints/042e067/ : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"654a8adce1ebec5b6e38e0b614884d8b6e03185a","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-16 20:31:44.000000000","message":"Uploaded patch set 84: Patch Set 83 was rebased.","accounts_in_message":[],"_revision_number":84},{"id":"aef8e6eeb7f52eb31a1934a7a30c6c042bcd7a91","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-16 20:48:19.000000000","message":"Patch Set 84:\n\n* pci-test http://52.27.155.124/pci/193135/84 : SUCCESS","accounts_in_message":[],"_revision_number":84},{"id":"fafaaad1ad6b50ae02341fd112816d8258be3674","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-16 20:50:34.000000000","message":"Patch Set 84: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/84/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8e4eae71e3894feaa586dd58d67808dc/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/84/check/real-db-upgrade_nova_mysql_user_001:th-mysql/df5492cb490f49f08bf4970080110926/index.html : SUCCESS in 14m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/84/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/862043831c744641ab22690a1fcf1bbc/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/84/check/real-db-upgrade_nova_percona_user_001:th-percona/cdc923f933964c689acad4644ee00785/index.html : SUCCESS in 15m 51s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/84/check/real-db-upgrade_nova_mysql_user_002:th-mysql/62f9a3a8dfad43a98c3923ea73cc8ba0/index.html : SUCCESS in 12m 30s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/84/check/real-db-upgrade_nova_percona_user_002:th-percona/be74b9cbc68a45aea6df059693803c8d/index.html : SUCCESS in 14m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":84},{"id":"47b00f8ecd4419fa7553edb196ee413c397c6d97","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 21:27:11.000000000","message":"Patch Set 84:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/84/check/check-dsvm-tempest-vz7-exe-minimal/8d83bbe : FAILURE in 53m 35s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":84},{"id":"71d47f1481c8b2d43507d21c542b5ab88ea89c24","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-16 21:42:34.000000000","message":"Patch Set 84:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/84/check/check-ibm-tempest-dsvm-full/dc0ee34/ : SUCCESS in 1h 09m 23s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"5fba38c757d867f3f29615038017f3716f988de8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-16 21:43:05.000000000","message":"Patch Set 84: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/84/3152/index.html : SUCCESS in 01 h 09 m 30 s","accounts_in_message":[],"_revision_number":84},{"id":"1b78282fa6aa5d0f2fdb1fc2d5d1c595b9f6e330","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-16 21:56:45.000000000","message":"Patch Set 84: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/84/check/dsvm-tempest-xen/6f54897 : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":84},{"id":"16ce371bed379115a4569b98a349a3736734ee8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-16 22:03:46.000000000","message":"Patch Set 84: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/84/check/gate-nova-pep8/5fae5cc/ : SUCCESS in 9m 45s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/84/check/gate-nova-docs/6093aee//doc/build/html/ : SUCCESS in 14m 24s\n- gate-nova-python27 http://logs.openstack.org/35/193135/84/check/gate-nova-python27/fadd272/ : SUCCESS in 14m 31s\n- gate-nova-python34 http://logs.openstack.org/35/193135/84/check/gate-nova-python34/a478fd5/ : SUCCESS in 29m 42s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-full/5d94292/ : SUCCESS in 1h 11m 39s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-postgres-full/f62e454/ : SUCCESS in 1h 08m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-neutron-full/73101a6/ : SUCCESS in 1h 27m 14s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/84/check/gate-grenade-dsvm/c10367a/ : SUCCESS in 44m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-large-ops/382226e/ : SUCCESS in 30m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-neutron-large-ops/4e6f5ac/ : SUCCESS in 34m 48s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/84/check/gate-nova-releasenotes/e8975ba//releasenotes/build/html/ : SUCCESS in 8m 44s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/84/check/gate-nova-tox-functional/7f7ee70/ : SUCCESS in 18m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/84/check/gate-grenade-dsvm-multinode/d5e4a10/ : SUCCESS in 49m 13s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0319951/ : SUCCESS in 44m 04s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-nova-v20-api/7661077/ : SUCCESS in 47m 29s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-nova-v20-api-legacy/35b1f24/ : SUCCESS in 39m 49s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-cells/359070f/ : SUCCESS in 52m 15s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-full-ceph/97291d9/ : SUCCESS in 1h 01m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/18e4e31/ : SUCCESS in 1h 12m 30s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/84/check/gate-tempest-dsvm-multinode-full/3778cc5/ : SUCCESS in 59m 53s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/84/check/gate-compute-api-guide/037270e//api-guide/build/html/ : SUCCESS in 8m 32s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/84/check/gate-nova-python27-constraints/beaada0/ : SUCCESS in 15m 28s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/84/check/gate-nova-python34-constraints/ab82404/ : SUCCESS in 28m 26s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/84/check/gate-nova-docs-constraints/0c5741b//doc/build/html/ : SUCCESS in 14m 57s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/84/check/gate-nova-pep8-constraints/b9e05da/ : SUCCESS in 10m 10s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/84/check/gate-nova-tox-constraints-functional-constraints/455b115/ : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"fdf3524c4b558a6bba1c627ca2b8cc66abd679cd","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-16 23:02:46.000000000","message":"Patch Set 84:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/84/ : FAILURE in 2h 28m 42s","accounts_in_message":[],"_revision_number":84},{"id":"7ef3669fab851be2a8890746153e0134014442be","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-17 21:06:39.000000000","message":"Patch Set 83:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/83/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":83},{"id":"50d51d39888815b568ef5c4f10175514d7792e34","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 00:50:05.000000000","message":"Patch Set 85: Patch Set 84 was rebased","accounts_in_message":[],"_revision_number":84},{"id":"157b7c7fa64caa97b0ee713f67ebd5831d9780bf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-19 00:59:35.000000000","message":"Patch Set 85:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/85/check/check-ibm-tempest-dsvm-full/c1daa07/ : FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":85},{"id":"8b40738ce483fb0684f8a74bbe38e358b57dc9ec","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-19 01:23:35.000000000","message":"Patch Set 85: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/85/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5ea57fe79da14d399e01834fc264c9ac/index.html : SUCCESS in 15m 09s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/85/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4a2d78552b6e418dac0e7449fa05b628/index.html : SUCCESS in 18m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/85/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1e134abf72fe4d1cb6506c72a15af969/index.html : SUCCESS in 14m 39s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/85/check/real-db-upgrade_nova_percona_user_001:th-percona/462179b587834e28a210e0890a845263/index.html : SUCCESS in 17m 24s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/85/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e1ac9ac99a804b1da29f7030f28b3d2d/index.html : SUCCESS in 15m 02s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/85/check/real-db-upgrade_nova_percona_user_002:th-percona/48a3b412e3224469b577bf4ec8a3be04/index.html : SUCCESS in 15m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":85},{"id":"607ffbc20762a94ef8ecaf2a6b5953dd54319ce7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-19 01:46:39.000000000","message":"Patch Set 85:\n\n* pci-test http://52.27.155.124/pci/193135/85 : SUCCESS","accounts_in_message":[],"_revision_number":85},{"id":"9a178cbc6b343ee2effb2c51f3f27c8700ab6164","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-19 02:06:45.000000000","message":"Patch Set 85: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/85/3337/index.html : SUCCESS in 01 h 04 m 58 s","accounts_in_message":[],"_revision_number":85},{"id":"9b29c819c34d9d3de79b57e583c4d641e41e130c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-19 02:16:08.000000000","message":"Patch Set 85: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/85/check/dsvm-tempest-xen/c377d2c : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":85},{"id":"e1f708ee758ea3213af9572c1ec9d521051a0b6b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 02:25:54.000000000","message":"Patch Set 85: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/85/check/gate-nova-pep8/80f3cf1/ : SUCCESS in 8m 12s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/85/check/gate-nova-docs/dafbf7f//doc/build/html/ : SUCCESS in 15m 46s\n- gate-nova-python27 http://logs.openstack.org/35/193135/85/check/gate-nova-python27/2eb16b1/ : SUCCESS in 16m 07s\n- gate-nova-python34 http://logs.openstack.org/35/193135/85/check/gate-nova-python34/7d61fa2/ : SUCCESS in 29m 38s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-full/4dd01cc/ : SUCCESS in 55m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-postgres-full/bdce7cb/ : SUCCESS in 53m 17s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-neutron-full/2237aa1/ : SUCCESS in 1h 30m 59s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/85/check/gate-grenade-dsvm/71274cd/ : SUCCESS in 51m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-large-ops/9be9451/ : FAILURE in 57m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-neutron-large-ops/686f385/ : SUCCESS in 34m 37s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/85/check/gate-nova-releasenotes/1d0cf31//releasenotes/build/html/ : SUCCESS in 7m 41s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/85/check/gate-nova-tox-functional/30222ca/ : SUCCESS in 16m 30s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/85/check/gate-grenade-dsvm-multinode/8ba4e52/ : SUCCESS in 49m 21s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e037a46/ : SUCCESS in 39m 43s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-nova-v20-api/82de92a/ : SUCCESS in 1h 04m 23s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-nova-v20-api-legacy/5806332/ : SUCCESS in 40m 46s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-cells/593d8a9/ : SUCCESS in 50m 30s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-full-ceph/5949bb8/ : SUCCESS in 1h 01m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/77ca893/ : SUCCESS in 1h 14m 24s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/85/check/gate-tempest-dsvm-multinode-full/c822790/ : FAILURE in 1h 08m 08s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/85/check/gate-compute-api-guide/f933568//api-guide/build/html/ : SUCCESS in 8m 50s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/85/check/gate-nova-python27-constraints/d9c9254/ : SUCCESS in 12m 53s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/85/check/gate-nova-python34-constraints/f95df26/ : SUCCESS in 26m 28s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/85/check/gate-nova-docs-constraints/67dda99//doc/build/html/ : SUCCESS in 10m 39s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/85/check/gate-nova-pep8-constraints/a818219/ : SUCCESS in 7m 57s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/85/check/gate-nova-tox-constraints-functional-constraints/64f7e79/ : SUCCESS in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":85},{"id":"574956f3b9aade506b2f42649892a62b7baa9158","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-19 03:17:36.000000000","message":"Patch Set 85:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/85/ : FAILURE in 2h 25m 12s","accounts_in_message":[],"_revision_number":85},{"id":"3d2a5cbf0d715c56daf3b0ed955adb0861891dfd","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 05:48:44.000000000","message":"Uploaded patch set 86: Patch Set 85 was rebased.","accounts_in_message":[],"_revision_number":86},{"id":"577a2e5d24451218263a3aeede46925eb7548496","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-19 06:13:42.000000000","message":"Patch Set 86: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9b3c19b89c4e44f2a26ec40ddc94f02e/index.html : SUCCESS in 13m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_mysql_user_001:th-mysql/85fe1e54ad1b4ab2a78c6f4ac3a7ed94/index.html : SUCCESS in 14m 50s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/44bc138334134756b7dd1505451b706f/index.html : SUCCESS in 17m 09s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_percona_user_001:th-percona/042bea7e41a444caa3f0d5ac7a5402f9/index.html : SUCCESS in 18m 21s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0b6d9f11da3e4768ab1c873a56088637/index.html : SUCCESS in 12m 47s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_percona_user_002:th-percona/f43646afa5f945559fbd59ae63992b76/index.html : SUCCESS in 14m 40s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":86},{"id":"454e2bb5be10a8b6116fcfb17eddf45519487896","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-19 06:59:44.000000000","message":"Patch Set 86: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/86/3361/index.html : SUCCESS in 01 h 07 m 18 s","accounts_in_message":[],"_revision_number":86},{"id":"0a6678dfe133f3e0e9b568f92bef7c6bf1bdcb68","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-19 07:02:35.000000000","message":"Patch Set 86:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/86/check/check-ibm-tempest-dsvm-full/97c83b6/ : SUCCESS in 1h 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":86},{"id":"af4a081304b4cd35bcbedff7e8688a65ef17d0aa","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-19 07:25:04.000000000","message":"Patch Set 86: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/86/check/dsvm-tempest-xen/4e7bb2d : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":86},{"id":"b4c3155589beb8252bcaa08c88dd864422d959f2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 07:29:11.000000000","message":"Patch Set 86: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/86/check/gate-nova-pep8/8f943cd/ : SUCCESS in 10m 45s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/86/check/gate-nova-docs/c8a04e9//doc/build/html/ : SUCCESS in 16m 26s\n- gate-nova-python27 http://logs.openstack.org/35/193135/86/check/gate-nova-python27/1d3726d/ : SUCCESS in 14m 46s\n- gate-nova-python34 http://logs.openstack.org/35/193135/86/check/gate-nova-python34/a549ee3/ : SUCCESS in 28m 12s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-full/8bca60e/ : SUCCESS in 1h 37m 40s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-postgres-full/8f06c98/ : SUCCESS in 1h 34m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-neutron-full/8ce49f7/ : SUCCESS in 1h 36m 50s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/86/check/gate-grenade-dsvm/81643e1/ : SUCCESS in 45m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-large-ops/620fef9/ : SUCCESS in 32m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-neutron-large-ops/dd1b31a/ : SUCCESS in 31m 40s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/86/check/gate-nova-releasenotes/2ba22a0//releasenotes/build/html/ : SUCCESS in 9m 35s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/86/check/gate-nova-tox-functional/58fdaee/ : SUCCESS in 16m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/86/check/gate-grenade-dsvm-multinode/2ba8b2e/ : FAILURE in 54m 00s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/5b6ae9b/ : SUCCESS in 35m 45s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-nova-v20-api/5e1b604/ : SUCCESS in 52m 44s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-nova-v20-api-legacy/0fdf442/ : SUCCESS in 41m 59s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-cells/ee0fa68/ : SUCCESS in 55m 01s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-full-ceph/1aa676a/ : SUCCESS in 1h 04m 52s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/1ac73e4/ : SUCCESS in 1h 07m 18s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-multinode-full/f7d33a8/ : FAILURE in 1h 15m 21s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/86/check/gate-compute-api-guide/9c0109e//api-guide/build/html/ : SUCCESS in 9m 22s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-python27-constraints/9d1c8cb/ : SUCCESS in 12m 29s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-python34-constraints/3cc208c/ : SUCCESS in 28m 14s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/86/check/gate-nova-docs-constraints/cf60b3a//doc/build/html/ : SUCCESS in 14m 43s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-pep8-constraints/9613c8e/ : SUCCESS in 8m 03s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-tox-constraints-functional-constraints/232800e/ : SUCCESS in 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":86},{"id":"01df985d59632caf848a428bd4726faa1dd116f6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-19 07:36:21.000000000","message":"Patch Set 86:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/86/ : SUCCESS in 1h 45m 16s","accounts_in_message":[],"_revision_number":86},{"id":"816d2228a8e1f310c9a74009514f0a20f4372fd2","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-01-19 07:41:16.000000000","message":"Patch Set 86:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-multinode: https://bugs.launchpad.net/bugs/1463631\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":86},{"id":"dd42a691f311fa5d489715c018cdb9def05273ab","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 07:54:58.000000000","message":"Patch Set 86:\n\nrecheck bug 1463631","accounts_in_message":[],"_revision_number":86},{"id":"3e15361cc9e03e4c59444fa63c1190d23030abcf","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-19 08:26:32.000000000","message":"Patch Set 86:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f2d2eded84044232821700eb3caad721/index.html : SUCCESS in 14m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_mysql_user_001:th-mysql/929e9d66089d49fe8ff32f2cd5265321/index.html : SUCCESS in 14m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/15a9048b0625458487be0b21d05db3e8/index.html : SUCCESS in 12m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_percona_user_001:th-percona/341b9cc865044a33a4fe804030ae4b59/index.html : SUCCESS in 16m 14s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c54bfc33be64347872dd3d930939e9f/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/86/check/real-db-upgrade_nova_percona_user_002:th-percona/bd8ea3b5159048bca4a3d285ed8b4b32/index.html : SUCCESS in 15m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":86},{"id":"f471dbd9b90e602f999ec78097128fadd9e2c439","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-19 09:26:37.000000000","message":"Patch Set 86:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/86/check/dsvm-tempest-xen/ea35670 : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":86},{"id":"25ce35f8bac637dd48bd584729ec8dcb866a7d51","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 09:36:31.000000000","message":"Patch Set 86: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/86/check/gate-nova-pep8/69b980c/ : SUCCESS in 9m 50s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/86/check/gate-nova-docs/e9d9cb2//doc/build/html/ : SUCCESS in 11m 51s\n- gate-nova-python27 http://logs.openstack.org/35/193135/86/check/gate-nova-python27/0bf6b96/ : SUCCESS in 12m 59s\n- gate-nova-python34 http://logs.openstack.org/35/193135/86/check/gate-nova-python34/70f8264/ : SUCCESS in 29m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-full/f609041/ : SUCCESS in 50m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-postgres-full/31c8ed5/ : SUCCESS in 49m 33s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-neutron-full/2fe51d2/ : SUCCESS in 1h 38m 27s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/86/check/gate-grenade-dsvm/cfd3a8f/ : SUCCESS in 50m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-large-ops/6e183d4/ : SUCCESS in 39m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-neutron-large-ops/4b7562e/ : SUCCESS in 51m 28s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/86/check/gate-nova-releasenotes/7795a6c//releasenotes/build/html/ : SUCCESS in 7m 04s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/86/check/gate-nova-tox-functional/7e66f55/ : SUCCESS in 17m 41s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/86/check/gate-grenade-dsvm-multinode/3bd33b1/ : SUCCESS in 51m 50s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/86cd227/ : SUCCESS in 46m 02s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-nova-v20-api/c0051c0/ : SUCCESS in 47m 22s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-nova-v20-api-legacy/2c6f13a/ : SUCCESS in 47m 23s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-cells/21344b7/ : SUCCESS in 52m 19s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-full-ceph/7b6c507/ : SUCCESS in 1h 11m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/76ce9af/ : SUCCESS in 1h 01m 22s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/86/check/gate-tempest-dsvm-multinode-full/d0f23b3/ : FAILURE in 1h 06m 18s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/86/check/gate-compute-api-guide/16ad4d3//api-guide/build/html/ : SUCCESS in 6m 55s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-python27-constraints/7bd3d77/ : SUCCESS in 12m 01s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-python34-constraints/a1c437b/ : SUCCESS in 25m 36s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/86/check/gate-nova-docs-constraints/4647074//doc/build/html/ : SUCCESS in 12m 08s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-pep8-constraints/7477914/ : SUCCESS in 9m 07s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/86/check/gate-nova-tox-constraints-functional-constraints/d83a532/ : SUCCESS in 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":86},{"id":"79faed402b8b73abc174b2f807a6946cd36b8e7c","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2016-01-19 09:38:07.000000000","message":"Patch Set 86:\n\n(2 comments)\n\nSo to add this support in any other driver we will need:-\n1. virt_driver.capabilities\n2. connection-info check","accounts_in_message":[],"_revision_number":86},{"id":"64f1e3f5265d48ad610a3e0b2ad4b559d6a68aa6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-19 09:53:09.000000000","message":"Patch Set 86:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/86/ : FAILURE in 1h 56m 46s","accounts_in_message":[],"_revision_number":86},{"id":"70bbb3a0944686cd04c7251815d6f049256ff068","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-19 11:29:46.000000000","message":"Patch Set 86:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/86/Nova-ML2-Sriov : FAILURE in 22m 32s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":86},{"id":"cc6e7cac2b8b940f31a8fc407b65bdf7b057b57c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 11:44:53.000000000","message":"Patch Set 86:\n\n(2 comments)","accounts_in_message":[],"_revision_number":86},{"id":"39ab1078131ea48fc8b64e51778238c8b6e539a8","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 18:41:00.000000000","message":"Uploaded patch set 87: Patch Set 86 was rebased.","accounts_in_message":[],"_revision_number":87},{"id":"91411a1ec2e93b1627d3520f0f1d4f744a1ea2b8","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 18:46:43.000000000","message":"Uploaded patch set 88: Patch Set 87 was rebased.","accounts_in_message":[],"_revision_number":88},{"id":"167031bf70f7846839c756556850468d08a04b3f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-19 20:02:05.000000000","message":"Patch Set 88:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/88/check/check-ibm-tempest-dsvm-full/ae79163/ : SUCCESS in 1h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"61904593d92cd4b471c1114c541125c2d357cdaa","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-19 20:12:23.000000000","message":"Patch Set 88: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/88/3506/index.html : SUCCESS in 01 h 14 m 05 s","accounts_in_message":[],"_revision_number":88},{"id":"23968538874ec1ac3e519de2007b3b6e0064478f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-19 20:26:11.000000000","message":"Patch Set 88: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/88/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/37719f6afdf843b98a6ab8ccb58e2f26/index.html : SUCCESS in 15m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/88/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e72258c1a830435588df03febc16e745/index.html : SUCCESS in 16m 26s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/88/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/420e469170a248cab5bd15abfd1b24dc/index.html : SUCCESS in 15m 22s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/88/check/real-db-upgrade_nova_percona_user_001:th-percona/9384512f9b834e789e9b3c010857d887/index.html : SUCCESS in 14m 29s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/88/check/real-db-upgrade_nova_mysql_user_002:th-mysql/60c4aa5027004f2e88bd78c965fca16d/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/88/check/real-db-upgrade_nova_percona_user_002:th-percona/2679684dacc14fee8441d2f453a604a8/index.html : SUCCESS in 12m 42s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":88},{"id":"f6dea5fdbce40117b80a5165866ece9930df3607","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-19 20:47:53.000000000","message":"Uploaded patch set 89: Patch Set 88 was rebased.","accounts_in_message":[],"_revision_number":89},{"id":"61a59cef6124a59fc144eed21e8d51ff728bc7a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-19 22:04:05.000000000","message":"Patch Set 89:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/89/check/check-ibm-tempest-dsvm-full/69c641a/ : SUCCESS in 1h 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"6a192177a315d5a7551a9ae663873c5dc6add499","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-19 22:04:06.000000000","message":"Patch Set 89: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/89/3525/index.html : SUCCESS in 01 h 12 m 14 s","accounts_in_message":[],"_revision_number":89},{"id":"7e99af23b3059bd574c06ceaa148e5bd5109e0d2","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-19 22:52:28.000000000","message":"Patch Set 89: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/89/check/dsvm-tempest-xen/edb1146 : FAILURE in 1h 31m 23s","accounts_in_message":[],"_revision_number":89},{"id":"fd593a535589b1a37a2c14273ec89332f36a5391","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-19 23:14:47.000000000","message":"Patch Set 89: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/89/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4e16c6398f464a27b5a0b800d2dcf7d5/index.html : SUCCESS in 15m 21s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/89/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0dd50101ae8e4a8a87f217048bb313ff/index.html : SUCCESS in 17m 01s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/89/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d56aa486320a4b3eb710cd2a857b61d5/index.html : SUCCESS in 15m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/89/check/real-db-upgrade_nova_percona_user_001:th-percona/c68a2dd7e37e4e72bfaabfb310956f47/index.html : SUCCESS in 16m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/89/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a126436565ce4072921084e1c3f1afa9/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/89/check/real-db-upgrade_nova_percona_user_002:th-percona/3eebe62c207644909e6fa46cfb62bab4/index.html : SUCCESS in 15m 42s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":89},{"id":"2705484f28062c0ee3a39c3a78653f6120812f8b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 23:44:23.000000000","message":"Patch Set 89: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/89/check/gate-nova-pep8/ffc504e/ : SUCCESS in 9m 06s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/89/check/gate-nova-docs/97927b8//doc/build/html/ : SUCCESS in 14m 04s\n- gate-nova-python27 http://logs.openstack.org/35/193135/89/check/gate-nova-python27/5887aa0/ : SUCCESS in 15m 59s\n- gate-nova-python34 http://logs.openstack.org/35/193135/89/check/gate-nova-python34/6927edd/ : SUCCESS in 29m 52s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-full/62d2f0e/ : SUCCESS in 1h 13m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-postgres-full/04de269/ : SUCCESS in 1h 06m 00s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-neutron-full/fbe7155/ : SUCCESS in 1h 41m 41s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/89/check/gate-grenade-dsvm/49387e2/ : SUCCESS in 57m 58s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-large-ops/75fafc6/ : SUCCESS in 48m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-neutron-large-ops/6602b5b/ : SUCCESS in 44m 46s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/89/check/gate-nova-releasenotes/66c7470//releasenotes/build/html/ : SUCCESS in 11m 15s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/89/check/gate-nova-tox-functional/65f2450/ : SUCCESS in 17m 07s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/89/check/gate-grenade-dsvm-multinode/49c5e32/ : SUCCESS in 55m 56s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/36971d1/ : SUCCESS in 44m 58s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-nova-v20-api/336ec46/ : SUCCESS in 45m 30s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-nova-v20-api-legacy/6c3a64e/ : SUCCESS in 47m 24s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-cells/416ada3/ : SUCCESS in 57m 51s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-full-ceph/a36837e/ : SUCCESS in 53m 25s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/f4cb97b/ : SUCCESS in 1h 03m 40s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/89/check/gate-tempest-dsvm-multinode-full/a4054dd/ : SUCCESS in 1h 08m 23s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/89/check/gate-compute-api-guide/b314678//api-guide/build/html/ : SUCCESS in 10m 02s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/89/check/gate-nova-python27-constraints/05d80ca/ : SUCCESS in 16m 03s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/89/check/gate-nova-python34-constraints/37a69e7/ : SUCCESS in 29m 22s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/89/check/gate-nova-docs-constraints/bdc462e//doc/build/html/ : SUCCESS in 15m 07s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/89/check/gate-nova-pep8-constraints/267ba5f/ : SUCCESS in 10m 08s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/89/check/gate-nova-tox-constraints-functional-constraints/8731345/ : SUCCESS in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"b7ab32948d360ddccd7d003de7ae2310ef1d8168","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-20 00:03:35.000000000","message":"Patch Set 89:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/89/check/check-dsvm-tempest-vz7-exe-minimal/aa602da : SUCCESS in 1h 33m 30s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":89},{"id":"11b267d97d0ef2d1de619eaf9f0cd9b87140bb6b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-20 00:32:23.000000000","message":"Patch Set 89:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/89/ : FAILURE in 3h 09m 03s","accounts_in_message":[],"_revision_number":89},{"id":"f5f2687c54174b45ca47f434537332135135a45d","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-20 07:11:29.000000000","message":"Patch Set 89:\n\nxen: recheck","accounts_in_message":[],"_revision_number":89},{"id":"8e2b5fafad030e0ac6ce6ce6a4d7e5c17fdcb042","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-20 10:02:08.000000000","message":"Patch Set 89:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/89/check/dsvm-tempest-xen/2dd8759 : FAILURE in 2h 49m 25s","accounts_in_message":[],"_revision_number":89},{"id":"77ccc346ea27325447c34ac023b1285d97d5ff52","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-20 10:30:49.000000000","message":"Patch Set 89:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/89/Nova-ML2-Sriov : FAILURE in 31m 19s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":89},{"id":"82f787c963adc81499055310c3c76387df705946","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-20 18:09:01.000000000","message":"Uploaded patch set 90: Patch Set 89 was rebased.","accounts_in_message":[],"_revision_number":90},{"id":"1dd34be320c86190a3f55f9b9196e0e19a4ee240","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-20 18:09:28.000000000","message":"Patch Set 90:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":90},{"id":"8641dc90d256db10b040234d27a0bcda1d041e81","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-20 18:09:28.000000000","message":"Patch Set 90:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":90},{"id":"a48f6585ee1c455f8402ecd91a5dc44a9781df85","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-20 18:09:28.000000000","message":"Patch Set 90: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":90},{"id":"8c4587506339e092fe32a668ab52cb197a9664e2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-20 18:09:50.000000000","message":"Patch Set 90:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":90},{"id":"291b85c680c1d5cf5f908c8358e00a49b604ef05","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-20 19:26:27.000000000","message":"Patch Set 90: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/90/3663/index.html : SUCCESS in 01 h 12 m 33 s","accounts_in_message":[],"_revision_number":90},{"id":"8fd02e71f3c79387dea944e86f9a0680cf178afc","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-20 21:05:25.000000000","message":"Uploaded patch set 91: Patch Set 90 was rebased.","accounts_in_message":[],"_revision_number":91},{"id":"79daae3ab073b3f843e0d00a7e1562e6ff05db11","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-20 21:18:20.000000000","message":"Uploaded patch set 92: Patch Set 91 was rebased.","accounts_in_message":[],"_revision_number":92},{"id":"ab635c8552950dda4dbf9545d54f2ea43bb47f9d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-20 21:44:11.000000000","message":"Patch Set 92:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/92/Nova-ML2-Sriov : FAILURE in 22m 07s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":92},{"id":"80f83f8f1e94f4186ea0a0d85e72e73b1f8a4297","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-20 22:23:23.000000000","message":"Patch Set 92:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/92/check/check-dsvm-tempest-vz7-exe-minimal/e6c6bd1 : FAILURE in 1h 03m 15s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":92},{"id":"a5e928cba866f6e67e462b129e6b93c8fec53974","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-20 22:34:17.000000000","message":"Patch Set 92:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/92/check/check-ibm-tempest-dsvm-full/43a7a7a/ : SUCCESS in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"aa9b15a4cc2a554c3e5be7ffed67b03d9c7f2c55","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-20 22:42:08.000000000","message":"Patch Set 92: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/92/3686/index.html : SUCCESS in 01 h 03 m 57 s","accounts_in_message":[],"_revision_number":92},{"id":"ae1f88386b23b7cc84e41f999df96dd9c0a7987e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-20 23:01:35.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/92/check/gate-nova-pep8/4056186/ : SUCCESS in 9m 22s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/92/check/gate-nova-docs/5dc410f//doc/build/html/ : SUCCESS in 13m 37s\n- gate-nova-python27 http://logs.openstack.org/35/193135/92/check/gate-nova-python27/dd179ad/ : SUCCESS in 13m 40s\n- gate-nova-python34 http://logs.openstack.org/35/193135/92/check/gate-nova-python34/ac439e0/ : SUCCESS in 28m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-full/9044816/ : SUCCESS in 1h 04m 01s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-postgres-full/734e0f3/ : SUCCESS in 47m 06s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-neutron-full/6c3da63/ : SUCCESS in 1h 33m 34s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/92/check/gate-grenade-dsvm/06d175f/ : SUCCESS in 44m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-large-ops/c437248/ : SUCCESS in 47m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-neutron-large-ops/4f3dae8/ : SUCCESS in 31m 13s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/92/check/gate-nova-releasenotes/5d96a14//releasenotes/build/html/ : SUCCESS in 8m 52s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/92/check/gate-nova-tox-functional/bc8bcc8/ : SUCCESS in 16m 26s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/92/check/gate-grenade-dsvm-multinode/6c2788d/ : SUCCESS in 1h 04m 03s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3ac1bee/ : SUCCESS in 49m 31s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-nova-v20-api/89ecccc/ : SUCCESS in 39m 49s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-nova-v20-api-legacy/b42e283/ : SUCCESS in 36m 12s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-cells/d29819e/ : SUCCESS in 1h 03m 50s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-full-ceph/ac230a6/ : SUCCESS in 1h 10m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/af98cbc/ : SUCCESS in 49m 31s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/92/check/gate-tempest-dsvm-multinode-full/9ba08e0/ : SUCCESS in 1h 24m 15s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/92/check/gate-compute-api-guide/89f88f7//api-guide/build/html/ : SUCCESS in 9m 11s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/92/check/gate-nova-python27-constraints/a5bec3d/ : SUCCESS in 15m 13s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/92/check/gate-nova-python34-constraints/ead8134/ : SUCCESS in 28m 16s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/92/check/gate-nova-docs-constraints/3a9cf2e//doc/build/html/ : SUCCESS in 14m 44s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/92/check/gate-nova-pep8-constraints/635ee81/ : SUCCESS in 10m 05s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/92/check/gate-nova-tox-constraints-functional-constraints/403dce5/ : SUCCESS in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"aade2538f62355825ebb1ce656b9c98be4b322dc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-20 23:05:35.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/92/check/dsvm-tempest-xen/2b7dedc : SUCCESS in 1h 36m 53s","accounts_in_message":[],"_revision_number":92},{"id":"84a045ec628ddf3e99168eb252e743818793c994","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 01:53:03.000000000","message":"Patch Set 92:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/92/ : FAILURE in 2h 48m 53s","accounts_in_message":[],"_revision_number":92},{"id":"546c575af1eb5448c7bac6bcde052a56fc04b13c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-21 14:11:15.000000000","message":"Uploaded patch set 93.","accounts_in_message":[],"_revision_number":93},{"id":"4263f7a3c3bd6f495d946cd17c604afc89ef4a78","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-21 15:19:53.000000000","message":"Patch Set 93: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/93/3792/index.html : SUCCESS in 01 h 05 m 34 s","accounts_in_message":[],"_revision_number":93},{"id":"3db5b82664cde5414a4ba18924cc60815e9623d4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 15:20:25.000000000","message":"Patch Set 93:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/93/ : FAILURE in 23m 21s","accounts_in_message":[],"_revision_number":93},{"id":"51f631083865670de09f50cedaaf3ae8b6d9526f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 15:21:49.000000000","message":"Patch Set 93:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/93/check/check-dsvm-tempest-vz7-exe-minimal/eed4fbd : FAILURE in 1h 06m 22s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":93},{"id":"6554f3efadfd6b848b4da4be236318dd461650be","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-21 15:44:08.000000000","message":"Patch Set 93:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/93/Nova-ML2-Sriov : FAILURE in 23m 19s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":93},{"id":"ff084a1c427478f41d3c51944a86b63aae72c7fc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 16:11:22.000000000","message":"Patch Set 93:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":93},{"id":"09a40d8a762340b3dadde2c4659723d8e09273f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-21 16:30:51.000000000","message":"Patch Set 93: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/93/check/gate-nova-pep8/81c5271/ : SUCCESS in 10m 05s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/93/check/gate-nova-docs/9672bc0//doc/build/html/ : SUCCESS in 14m 24s\n- gate-nova-python27 http://logs.openstack.org/35/193135/93/check/gate-nova-python27/c1ca0bf/ : SUCCESS in 14m 22s\n- gate-nova-python34 http://logs.openstack.org/35/193135/93/check/gate-nova-python34/849ee3a/ : SUCCESS in 29m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-full/c7b4655/ : SUCCESS in 1h 05m 34s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-postgres-full/9d0e8b9/ : SUCCESS in 53m 48s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-neutron-full/76c235c/ : SUCCESS in 1h 23m 58s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/93/check/gate-grenade-dsvm/1f1f5b0/ : FAILURE in 53m 01s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-large-ops/36f2c1b/ : SUCCESS in 34m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-neutron-large-ops/b5deaf4/ : SUCCESS in 42m 02s\n- gate-nova-releasenotes http://docs-draft.openstack.org/35/193135/93/check/gate-nova-releasenotes/20507ce//releasenotes/build/html/ : SUCCESS in 9m 10s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/93/check/gate-nova-tox-functional/47023f9/ : SUCCESS in 17m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/93/check/gate-grenade-dsvm-multinode/eb76333/ : SUCCESS in 55m 21s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ae7db8c/ : SUCCESS in 51m 39s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-nova-v20-api/42a8ab9/ : SUCCESS in 49m 35s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-nova-v20-api-legacy/e8899a2/ : SUCCESS in 50m 18s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-cells/639fd0b/ : SUCCESS in 53m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-full-ceph/eaf8774/ : SUCCESS in 1h 23m 04s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/1ef8233/ : SUCCESS in 1h 00m 33s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/93/check/gate-tempest-dsvm-multinode-full/4826b75/ : FAILURE in 1h 50m 19s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/93/check/gate-compute-api-guide/4749dc0//api-guide/build/html/ : SUCCESS in 9m 25s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/93/check/gate-nova-python27-constraints/460b660/ : SUCCESS in 18m 03s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/93/check/gate-nova-python34-constraints/a1a801a/ : SUCCESS in 28m 38s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/93/check/gate-nova-docs-constraints/4881359//doc/build/html/ : SUCCESS in 16m 49s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/93/check/gate-nova-pep8-constraints/d2ad78e/ : SUCCESS in 10m 24s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/93/check/gate-nova-tox-constraints-functional-constraints/2a2e0ec/ : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"c881dfba054aed5e33b08817bcd2c61d37e1fe98","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-21 16:51:16.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/93/check/dsvm-tempest-xen/c4f6ed8 : SUCCESS in 2h 33m 57s","accounts_in_message":[],"_revision_number":93},{"id":"9ca40e73a3ae13e846df339bbc034952346a721a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-21 17:35:09.000000000","message":"Patch Set 93:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/93/check/check-ibm-tempest-dsvm-full/115aeef/ : SUCCESS in 1h 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"d6b24d443654d954cb0bedf854d9922035131c76","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 18:03:26.000000000","message":"Patch Set 93:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/93/ : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":93},{"id":"179931aa095216c3dc0b109d6a4f8c43b411183c","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-01-21 18:17:52.000000000","message":"Uploaded patch set 94.","accounts_in_message":[],"_revision_number":94},{"id":"5fe1c5e610f4c0a061553688feb718cd4cb278c7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 19:12:24.000000000","message":"Patch Set 94:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/35/193135/94/check/check-dsvm-tempest-vz7-exe-minimal/e931754 : SUCCESS in 52m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":94},{"id":"dd795ade4efd0add23e72e16d4e7e8651278794b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-21 19:22:55.000000000","message":"Patch Set 94:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/193135/94/Nova-ML2-Sriov : FAILURE in 23m 19s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":94},{"id":"a117d2cd07cc65f656eaa1b965153a6ff7f69d30","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-21 19:27:17.000000000","message":"Patch Set 94: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/35/193135/94/3818/index.html : SUCCESS in 01 h 03 m 54 s","accounts_in_message":[],"_revision_number":94},{"id":"7f1a46c66a6fe456a27f85d7ad5bbab3bf98bde0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-21 19:29:27.000000000","message":"Patch Set 94: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/35/193135/94/check/dsvm-tempest-xen/86d85ff : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":94},{"id":"dd21184c5233421939c6a8cef1f88aa0fcf7291c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-21 19:41:13.000000000","message":"Patch Set 94:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/35/193135/94/check/check-ibm-tempest-dsvm-full/3f61b9c/ : FAILURE in 1h 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":94},{"id":"709288dd54eefdb784bfaf078b64757a7775e7d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-21 20:47:07.000000000","message":"Patch Set 94: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/35/193135/94/check/gate-nova-pep8/d1cee1b/ : SUCCESS in 9m 19s\n- gate-nova-docs http://docs-draft.openstack.org/35/193135/94/check/gate-nova-docs/a0b1352//doc/build/html/ : SUCCESS in 9m 59s\n- gate-nova-python27 http://logs.openstack.org/35/193135/94/check/gate-nova-python27/8010ebe/ : SUCCESS in 15m 07s\n- gate-nova-python34 http://logs.openstack.org/35/193135/94/check/gate-nova-python34/d8a1800/ : SUCCESS in 30m 22s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-full/c434a37/ : SUCCESS in 55m 11s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-postgres-full/d64986f/ : SUCCESS in 50m 57s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-neutron-full/a9a6be5/ : SUCCESS in 1h 18m 41s\n- gate-grenade-dsvm http://logs.openstack.org/35/193135/94/check/gate-grenade-dsvm/d596310/ : SUCCESS in 53m 55s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-large-ops/1764413/ : SUCCESS in 45m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-neutron-large-ops/18deb78/ : SUCCESS in 46m 45s\n- gate-nova-tox-functional http://logs.openstack.org/35/193135/94/check/gate-nova-tox-functional/ab56309/ : SUCCESS in 16m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/35/193135/94/check/gate-grenade-dsvm-multinode/1f965a7/ : SUCCESS in 53m 05s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/fde09f7/ : SUCCESS in 54m 31s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-nova-v20-api/8fda5b2/ : SUCCESS in 56m 02s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-nova-v20-api-legacy/985e271/ : SUCCESS in 53m 51s\n- gate-tempest-dsvm-cells http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-cells/f1817b2/ : SUCCESS in 1h 00m 05s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-full-ceph/7301e2d/ : SUCCESS in 52m 36s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/31a1041/ : SUCCESS in 53m 16s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/35/193135/94/check/gate-tempest-dsvm-multinode-full/521e701/ : FAILURE in 1h 04m 01s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/35/193135/94/check/gate-compute-api-guide/04a43c0//api-guide/build/html/ : SUCCESS in 8m 57s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/35/193135/94/check/gate-nova-python27-constraints/d9e497f/ : SUCCESS in 14m 52s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/35/193135/94/check/gate-nova-python34-constraints/dcc7206/ : SUCCESS in 29m 01s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/35/193135/94/check/gate-nova-docs-constraints/010a1c4//doc/build/html/ : SUCCESS in 14m 35s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/35/193135/94/check/gate-nova-pep8-constraints/9ec776a/ : SUCCESS in 10m 17s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/35/193135/94/check/gate-nova-tox-constraints-functional-constraints/eee5e75/ : SUCCESS in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":94},{"id":"78c469bdd4b809011ba2c6376d2886be4d9e927c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-21 22:58:33.000000000","message":"Patch Set 94:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/193135/94/ : FAILURE in 4h 21m 21s","accounts_in_message":[],"_revision_number":94},{"id":"000acc1c9248e921f28acdf93d058ca86953ec8a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-22 06:48:31.000000000","message":"Patch Set 93:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/93 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":93},{"id":"ee0d3d0bada551cd7e9b4823276b756503d821e6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-22 06:49:12.000000000","message":"Patch Set 94:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/35/193135/94 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":94},{"id":"0b42cfa7f2ad34aa9d1321f0774e65d20caa495e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-23 22:10:39.000000000","message":"Patch Set 94:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/193135/94/1572 : FAILURE","accounts_in_message":[],"_revision_number":94},{"id":"e7513928f4cf5b0e0fcbd79765d8c49e7c4e573f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-27 19:37:03.000000000","message":"Patch Set 94: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":94},{"id":"9039e05fff722ed2a22122d2cce7e410e1b0f298","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-28 23:18:52.000000000","message":"Patch Set 94: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/94/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/c4aabfd506f54f0290fd77c7d7ee7402/index.html : SUCCESS in 13m 47s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/94/check/real-db-upgrade_nova_mysql_user_001:th-mysql/247538d221b84c6bb49e2e7f25691969/index.html : SUCCESS in 16m 12s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/94/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/02c710fe6a3d490f8a4263e469d177cf/index.html : SUCCESS in 14m 22s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/94/check/real-db-upgrade_nova_percona_user_001:th-percona/5c60686325c147729a4735041a2e57f9/index.html : SUCCESS in 16m 37s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/35/193135/94/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6cdcdf60f6ab4878b001d0ae777aec5f/index.html : SUCCESS in 12m 54s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/35/193135/94/check/real-db-upgrade_nova_percona_user_002:th-percona/ea4c6131787648979fa432138e844c5a/index.html : SUCCESS in 15m 17s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":94},{"id":"483087c4d7a3bea89af222729904d62c23cc795c","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-02-02 14:49:29.000000000","message":"Abandoned\n\nrequested abandon on IRC.","accounts_in_message":[],"_revision_number":94},{"id":"f5830f3ef810b2c970e173f7752599fba1d17d46","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2016-02-02 15:21:37.000000000","message":"Patch Set 79:\n\nThis patch was merged with the compute changes as by the end of the discussions that patch had only a few new test cases remaining. The latest version is available here: https://review.openstack.org/#/c/267587","accounts_in_message":[],"_revision_number":79}],"current_revision_number":94,"current_revision":"a62d37c106473e172a8cd80270b848843af8ffab","revisions":{"48252f031bbf6c94544b4294b133c43bce6cca74":{"kind":"REWORK","_number":1,"created":"2015-06-18 13:45:12.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/1"}}},"commit":{"parents":[{"commit":"a244d37fc90a87d7fc5cc0600aca63cc50a28d5b","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a244d37fc90a87d7fc5cc0600aca63cc50a28d5b"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 13:45:02.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48252f031bbf6c94544b4294b133c43bce6cca74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48252f031bbf6c94544b4294b133c43bce6cca74"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a244d37fc90a87d7fc5cc0600aca63cc50a28d5b","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5bf7e2e354a23d5dda1f9fe7d08ebbec0a28a415":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2015-06-18 13:52:01.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/2"}}},"commit":{"parents":[{"commit":"e2b2bcf8fa59d42379c0a4dc9a5505659836c2c1","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e2b2bcf8fa59d42379c0a4dc9a5505659836c2c1"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 13:51:22.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bf7e2e354a23d5dda1f9fe7d08ebbec0a28a415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bf7e2e354a23d5dda1f9fe7d08ebbec0a28a415"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e2b2bcf8fa59d42379c0a4dc9a5505659836c2c1","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"10679248d421b2ce760e9e75b75ac201cd68ee15":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2015-06-19 08:51:44.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/3"}}},"commit":{"parents":[{"commit":"f8c938f7898f64848f246ef4f6c7d6388e458246","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8c938f7898f64848f246ef4f6c7d6388e458246"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-19 08:50:07.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10679248d421b2ce760e9e75b75ac201cd68ee15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10679248d421b2ce760e9e75b75ac201cd68ee15"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f8c938f7898f64848f246ef4f6c7d6388e458246","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5b88a2db0f6f89a63dc7857c1602a111150b653c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2015-06-24 16:07:58.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/4"}}},"commit":{"parents":[{"commit":"89e0324f040796f437f54ecb6a233bb96fb7b2f0","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89e0324f040796f437f54ecb6a233bb96fb7b2f0"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-24 15:55:07.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b88a2db0f6f89a63dc7857c1602a111150b653c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b88a2db0f6f89a63dc7857c1602a111150b653c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"89e0324f040796f437f54ecb6a233bb96fb7b2f0","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2844dc769345ea21321e1d0a12ad8a5f171fc458":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2015-07-03 12:02:14.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/5"}}},"commit":{"parents":[{"commit":"c9c79a291f068cd1343b491d1d89a9cf8a3c25dd","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c9c79a291f068cd1343b491d1d89a9cf8a3c25dd"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-07-03 11:45:19.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2844dc769345ea21321e1d0a12ad8a5f171fc458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2844dc769345ea21321e1d0a12ad8a5f171fc458"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c9c79a291f068cd1343b491d1d89a9cf8a3c25dd","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b89cf9671a8529108a37ba62c0f65855e200d6e4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2015-07-07 10:11:38.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/6"}}},"commit":{"parents":[{"commit":"4fd5eb4834b8c4138d3c108615c2ada73c4e9d12","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4fd5eb4834b8c4138d3c108615c2ada73c4e9d12"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-07-07 10:09:48.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b89cf9671a8529108a37ba62c0f65855e200d6e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b89cf9671a8529108a37ba62c0f65855e200d6e4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4fd5eb4834b8c4138d3c108615c2ada73c4e9d12","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0e2ee84d9126b9099c1c623e39456bc73a4e2b8c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2015-07-07 14:13:17.000000000","uploader":{"_account_id":13909,"name":"Chris Gacsal","email":"chris.gacsal@gmail.com","username":"chrisg"},"ref":"refs/changes/35/193135/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/7"}}},"commit":{"parents":[{"commit":"41c900c942e508e4b8c657398f3f4b22ad88e1dd","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/41c900c942e508e4b8c657398f3f4b22ad88e1dd"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-07-07 14:11:20.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0e2ee84d9126b9099c1c623e39456bc73a4e2b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0e2ee84d9126b9099c1c623e39456bc73a4e2b8c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"41c900c942e508e4b8c657398f3f4b22ad88e1dd","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"89a6a0dc5e38f8099a308ebfa81490586d6c979c":{"kind":"REWORK","_number":8,"created":"2015-08-27 13:20:47.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/193135/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/8"}}},"commit":{"parents":[{"commit":"cc0d3bfe5ac1ccc535960798a768cf9d733a1a87","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc0d3bfe5ac1ccc535960798a768cf9d733a1a87"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2015-08-27 13:12:21.000000000","tz":120},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89a6a0dc5e38f8099a308ebfa81490586d6c979c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89a6a0dc5e38f8099a308ebfa81490586d6c979c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc0d3bfe5ac1ccc535960798a768cf9d733a1a87","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"38a4f283aa3b95a019ae8a771b3f5e00641c39db":{"kind":"REWORK","_number":9,"created":"2015-11-29 18:11:41.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/9"}}},"commit":{"parents":[{"commit":"aa132d9250b4e4c51a516ec0b3ceb6eb87162c01","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa132d9250b4e4c51a516ec0b3ceb6eb87162c01"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-11-29 18:10:16.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38a4f283aa3b95a019ae8a771b3f5e00641c39db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38a4f283aa3b95a019ae8a771b3f5e00641c39db"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa132d9250b4e4c51a516ec0b3ceb6eb87162c01","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2ad4770d0e2db2df435794aa3d14b451b3cb471c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2015-11-29 19:14:08.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/10"}}},"commit":{"parents":[{"commit":"c3504c754b9060b909d4f8d170285bfe5ebce76d","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3504c754b9060b909d4f8d170285bfe5ebce76d"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-11-29 19:12:35.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ad4770d0e2db2df435794aa3d14b451b3cb471c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ad4770d0e2db2df435794aa3d14b451b3cb471c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c3504c754b9060b909d4f8d170285bfe5ebce76d","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f6e603159243c9a0942c1c43684a8d975a5d444b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2015-11-29 21:08:12.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/11"}}},"commit":{"parents":[{"commit":"55b732ca8edbe915c87ba189322955b6e90f083f","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55b732ca8edbe915c87ba189322955b6e90f083f"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-11-29 21:07:55.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6e603159243c9a0942c1c43684a8d975a5d444b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6e603159243c9a0942c1c43684a8d975a5d444b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"55b732ca8edbe915c87ba189322955b6e90f083f","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a0aa1c676a933bc4cc8c8f1c30b6f7e7cb77b0fc":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2015-12-04 21:38:08.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/12"}}},"commit":{"parents":[{"commit":"5de93c84fdd01dd495bffef0ff435fab6e38cba3","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5de93c84fdd01dd495bffef0ff435fab6e38cba3"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-04 21:37:09.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0aa1c676a933bc4cc8c8f1c30b6f7e7cb77b0fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0aa1c676a933bc4cc8c8f1c30b6f7e7cb77b0fc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de93c84fdd01dd495bffef0ff435fab6e38cba3","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3fa8bbf652de1d03d3904f636e23e5ff066c529d":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2015-12-05 21:23:00.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/13"}}},"commit":{"parents":[{"commit":"a189168101c5d856722acb409e594741ecc44889","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a189168101c5d856722acb409e594741ecc44889"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-05 21:22:37.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fa8bbf652de1d03d3904f636e23e5ff066c529d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fa8bbf652de1d03d3904f636e23e5ff066c529d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a189168101c5d856722acb409e594741ecc44889","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7bb19b227b0075552327d75cfcd826903e5f0d73":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2015-12-05 21:23:56.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/14"}}},"commit":{"parents":[{"commit":"1310d44bbe7594f07f2924812094a29f777f2070","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1310d44bbe7594f07f2924812094a29f777f2070"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-05 21:23:43.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bb19b227b0075552327d75cfcd826903e5f0d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bb19b227b0075552327d75cfcd826903e5f0d73"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1310d44bbe7594f07f2924812094a29f777f2070","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bd521f6bda18a6202fc1d68e629ca0f4b2578ae8":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2015-12-06 11:08:48.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/15"}}},"commit":{"parents":[{"commit":"4e5970cf10eb922e4507548905758d3982304ab6","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e5970cf10eb922e4507548905758d3982304ab6"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-06 11:08:27.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd521f6bda18a6202fc1d68e629ca0f4b2578ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd521f6bda18a6202fc1d68e629ca0f4b2578ae8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4e5970cf10eb922e4507548905758d3982304ab6","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d62c9b97e107c9d818a77e27f71d465aa5ee81d6":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2015-12-07 15:05:10.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/16"}}},"commit":{"parents":[{"commit":"f9f268acbf5a6fabc2b1d918b818d2037838d89e","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9f268acbf5a6fabc2b1d918b818d2037838d89e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-07 15:04:45.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d62c9b97e107c9d818a77e27f71d465aa5ee81d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d62c9b97e107c9d818a77e27f71d465aa5ee81d6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9f268acbf5a6fabc2b1d918b818d2037838d89e","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f2e0854d2e8e7d482c96bfb525423d8b22d71802":{"kind":"REWORK","_number":17,"created":"2015-12-07 15:23:05.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/17"}}},"commit":{"parents":[{"commit":"f9f268acbf5a6fabc2b1d918b818d2037838d89e","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9f268acbf5a6fabc2b1d918b818d2037838d89e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-07 15:22:45.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2e0854d2e8e7d482c96bfb525423d8b22d71802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2e0854d2e8e7d482c96bfb525423d8b22d71802"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9f268acbf5a6fabc2b1d918b818d2037838d89e","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"55f01a0e4a3bffd6293e5897d50567afd464edf0":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2015-12-07 17:45:00.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/18"}}},"commit":{"parents":[{"commit":"cf9e791307e756ba14fca7566d0d8ed9aba63169","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cf9e791307e756ba14fca7566d0d8ed9aba63169"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-07 17:44:42.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55f01a0e4a3bffd6293e5897d50567afd464edf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55f01a0e4a3bffd6293e5897d50567afd464edf0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cf9e791307e756ba14fca7566d0d8ed9aba63169","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"eaa240a1b7c470ac3de9f2286d1ad1006d5a5cc8":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2015-12-07 22:05:53.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/19"}}},"commit":{"parents":[{"commit":"0a60a79e094f216466f0422e48c0db55b2eaa051","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a60a79e094f216466f0422e48c0db55b2eaa051"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-07 22:01:18.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eaa240a1b7c470ac3de9f2286d1ad1006d5a5cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eaa240a1b7c470ac3de9f2286d1ad1006d5a5cc8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0a60a79e094f216466f0422e48c0db55b2eaa051","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":18,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8b7c997fe76f1e509f3e92505ed4b387cf5a331c":{"kind":"REWORK","_number":20,"created":"2015-12-08 13:36:40.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/20"}}},"commit":{"parents":[{"commit":"c6994a165f2a46165c0e15167871971348650004","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6994a165f2a46165c0e15167871971348650004"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-08 12:28:36.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b7c997fe76f1e509f3e92505ed4b387cf5a331c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b7c997fe76f1e509f3e92505ed4b387cf5a331c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c6994a165f2a46165c0e15167871971348650004","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":19,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e08272118c46a9776815026a03b3a0730c395130":{"kind":"REWORK","_number":21,"created":"2015-12-08 16:48:10.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/21"}}},"commit":{"parents":[{"commit":"c6994a165f2a46165c0e15167871971348650004","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6994a165f2a46165c0e15167871971348650004"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-08 16:21:51.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e08272118c46a9776815026a03b3a0730c395130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e08272118c46a9776815026a03b3a0730c395130"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c6994a165f2a46165c0e15167871971348650004","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":19,"change_status":"MERGED"}],"branch":"refs/heads/master"},"48160d845ae95d3ab8bd63437f53da04144b5b1b":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2015-12-09 14:20:31.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/22"}}},"commit":{"parents":[{"commit":"7802e7ad45573a7392151b03a21a56628a4f3ea2","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7802e7ad45573a7392151b03a21a56628a4f3ea2"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-09 14:17:30.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48160d845ae95d3ab8bd63437f53da04144b5b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48160d845ae95d3ab8bd63437f53da04144b5b1b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7802e7ad45573a7392151b03a21a56628a4f3ea2","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":20,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8ba10e2a61982d9a4b98e55b9975c33865996058":{"kind":"REWORK","_number":23,"created":"2015-12-09 19:27:21.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/23"}}},"commit":{"parents":[{"commit":"7802e7ad45573a7392151b03a21a56628a4f3ea2","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7802e7ad45573a7392151b03a21a56628a4f3ea2"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-09 19:26:46.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ba10e2a61982d9a4b98e55b9975c33865996058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ba10e2a61982d9a4b98e55b9975c33865996058"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7802e7ad45573a7392151b03a21a56628a4f3ea2","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":20,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a92cffeb3c2662aae06b1f1d6d0fbe18947928f4":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2015-12-09 19:35:17.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/24"}}},"commit":{"parents":[{"commit":"21e96bd49b9b5971c2ca6e5fd1fd5cc79bb99396","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/21e96bd49b9b5971c2ca6e5fd1fd5cc79bb99396"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-09 19:34:53.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a92cffeb3c2662aae06b1f1d6d0fbe18947928f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a92cffeb3c2662aae06b1f1d6d0fbe18947928f4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"21e96bd49b9b5971c2ca6e5fd1fd5cc79bb99396","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":21,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cd01a1c17a4335547a9fa14904136950f6aee746":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2015-12-11 09:21:12.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/25"}}},"commit":{"parents":[{"commit":"d16551dbe16897ce420fa496fccc4bcccf50077f","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d16551dbe16897ce420fa496fccc4bcccf50077f"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-11 09:20:58.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd01a1c17a4335547a9fa14904136950f6aee746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd01a1c17a4335547a9fa14904136950f6aee746"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d16551dbe16897ce420fa496fccc4bcccf50077f","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":22,"change_status":"MERGED"}],"branch":"refs/heads/master"},"feb9ea5c93278b896b4ed3c6b9b8fdb4bcb5b4e5":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2015-12-11 20:45:38.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/26"}}},"commit":{"parents":[{"commit":"b093fd5cf6aa9ca609b154100692fabccfdbc85a","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b093fd5cf6aa9ca609b154100692fabccfdbc85a"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-11 20:44:45.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/feb9ea5c93278b896b4ed3c6b9b8fdb4bcb5b4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/feb9ea5c93278b896b4ed3c6b9b8fdb4bcb5b4e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b093fd5cf6aa9ca609b154100692fabccfdbc85a","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":23,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a2b5ff41e33fce6f4e8511572919f1cfdf5f5905":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2015-12-13 11:11:05.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/27"}}},"commit":{"parents":[{"commit":"2fcd3adb59bbfeead809365e8fef0d12cb407cf2","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fcd3adb59bbfeead809365e8fef0d12cb407cf2"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-13 11:07:20.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2b5ff41e33fce6f4e8511572919f1cfdf5f5905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2b5ff41e33fce6f4e8511572919f1cfdf5f5905"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2fcd3adb59bbfeead809365e8fef0d12cb407cf2","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":24,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0c508b73de555f51f144fdc889bff53690bc6143":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2015-12-13 14:27:36.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/28"}}},"commit":{"parents":[{"commit":"58cbc2d8b58089464b55ea48264abbe3019dba30","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/58cbc2d8b58089464b55ea48264abbe3019dba30"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-13 13:28:37.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c508b73de555f51f144fdc889bff53690bc6143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c508b73de555f51f144fdc889bff53690bc6143"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"58cbc2d8b58089464b55ea48264abbe3019dba30","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":25,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ea541345ba42384866602da848618151f5e770f2":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2015-12-13 16:17:44.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/29"}}},"commit":{"parents":[{"commit":"d1a8c2a050297f84b143edf33b16a012866407d2","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1a8c2a050297f84b143edf33b16a012866407d2"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-13 16:17:37.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea541345ba42384866602da848618151f5e770f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea541345ba42384866602da848618151f5e770f2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d1a8c2a050297f84b143edf33b16a012866407d2","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":26,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a4c1c20a1e2c88f8164bc8b9a53bd79c9ec47c85":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2015-12-15 06:03:06.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/30"}}},"commit":{"parents":[{"commit":"6c4e274264c2e16f19e83d51b5734abb16ce3a85","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c4e274264c2e16f19e83d51b5734abb16ce3a85"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-15 05:50:36.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4c1c20a1e2c88f8164bc8b9a53bd79c9ec47c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4c1c20a1e2c88f8164bc8b9a53bd79c9ec47c85"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c4e274264c2e16f19e83d51b5734abb16ce3a85","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":27,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7826e2cc44b670ebf46e552efdf1261f5c87a717":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2015-12-15 12:52:13.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/31"}}},"commit":{"parents":[{"commit":"4186e750b08f47ab80a83d6844803ae0c20cee15","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4186e750b08f47ab80a83d6844803ae0c20cee15"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-15 12:52:06.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7826e2cc44b670ebf46e552efdf1261f5c87a717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7826e2cc44b670ebf46e552efdf1261f5c87a717"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4186e750b08f47ab80a83d6844803ae0c20cee15","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":28,"change_status":"MERGED"}],"branch":"refs/heads/master"},"405e1507ab04dea1fec48b3afc16a194865667e6":{"kind":"NO_CHANGE","_number":32,"created":"2015-12-16 13:06:32.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/32"}}},"commit":{"parents":[{"commit":"6c27cdd9f518affe0b993c4ea2384a1adac53f30","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c27cdd9f518affe0b993c4ea2384a1adac53f30"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-16 13:06:19.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/405e1507ab04dea1fec48b3afc16a194865667e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/405e1507ab04dea1fec48b3afc16a194865667e6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c27cdd9f518affe0b993c4ea2384a1adac53f30","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":29,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d44ae2857aaae742ac274e51efe65aca8504a3a1":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2015-12-16 13:10:04.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/33"}}},"commit":{"parents":[{"commit":"d80967882ebd2327c7ae76532c3a247e59fd2997","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d80967882ebd2327c7ae76532c3a247e59fd2997"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-16 13:09:48.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d44ae2857aaae742ac274e51efe65aca8504a3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d44ae2857aaae742ac274e51efe65aca8504a3a1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d80967882ebd2327c7ae76532c3a247e59fd2997","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":30,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c20abb6363b15d3b9854c33f31a398943c612aec":{"kind":"REWORK","_number":34,"created":"2015-12-16 15:34:01.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/34"}}},"commit":{"parents":[{"commit":"d80967882ebd2327c7ae76532c3a247e59fd2997","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d80967882ebd2327c7ae76532c3a247e59fd2997"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-16 14:34:54.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c20abb6363b15d3b9854c33f31a398943c612aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c20abb6363b15d3b9854c33f31a398943c612aec"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d80967882ebd2327c7ae76532c3a247e59fd2997","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":30,"change_status":"MERGED"}],"branch":"refs/heads/master"},"70e2720fd958bf6bdb9c6eb83da083bd47aedba8":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2015-12-16 16:13:26.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/35"}}},"commit":{"parents":[{"commit":"e87d4a4924fe638377f500df602a4cd1b3029a09","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e87d4a4924fe638377f500df602a4cd1b3029a09"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-16 16:13:19.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70e2720fd958bf6bdb9c6eb83da083bd47aedba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70e2720fd958bf6bdb9c6eb83da083bd47aedba8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e87d4a4924fe638377f500df602a4cd1b3029a09","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":31,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d2c61c185f71541e750ac847f07877637f6567a5":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2015-12-17 06:55:46.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/36"}}},"commit":{"parents":[{"commit":"eea4c78f4120423d9b5d9c4bd72cfd011300a160","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eea4c78f4120423d9b5d9c4bd72cfd011300a160"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-17 06:54:54.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\nREST API changes:\nhttps://review.openstack.org/153038\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2c61c185f71541e750ac847f07877637f6567a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2c61c185f71541e750ac847f07877637f6567a5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eea4c78f4120423d9b5d9c4bd72cfd011300a160","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":32,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0357ae7b1de9e8cfea084989289db2b1b2e39eff":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":37,"created":"2015-12-17 21:10:38.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/37"}}},"commit":{"parents":[{"commit":"6c1ed4cbf51d00bcbad60eb6924450880ccf174e","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c1ed4cbf51d00bcbad60eb6924450880ccf174e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-17 20:11:39.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0357ae7b1de9e8cfea084989289db2b1b2e39eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0357ae7b1de9e8cfea084989289db2b1b2e39eff"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c1ed4cbf51d00bcbad60eb6924450880ccf174e","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":33,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b8d1d09379378bfba49285d6df296811030f4048":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2015-12-23 05:57:54.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/38"}}},"commit":{"parents":[{"commit":"b032a4430eea027d7807610673888fef588a7e5f","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b032a4430eea027d7807610673888fef588a7e5f"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-22 22:38:27.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8d1d09379378bfba49285d6df296811030f4048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8d1d09379378bfba49285d6df296811030f4048"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b032a4430eea027d7807610673888fef588a7e5f","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":34,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d452dc252afb02a07d45f4847220ce1d37823884":{"kind":"REWORK","_number":39,"created":"2015-12-24 12:51:08.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/39"}}},"commit":{"parents":[{"commit":"b032a4430eea027d7807610673888fef588a7e5f","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b032a4430eea027d7807610673888fef588a7e5f"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-24 12:33:03.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d452dc252afb02a07d45f4847220ce1d37823884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d452dc252afb02a07d45f4847220ce1d37823884"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b032a4430eea027d7807610673888fef588a7e5f","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":34,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ca24b7a5e39915463eb475c6daaa31857e89b78b":{"kind":"REWORK","_number":40,"created":"2015-12-25 10:38:50.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/40"}}},"commit":{"parents":[{"commit":"b032a4430eea027d7807610673888fef588a7e5f","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b032a4430eea027d7807610673888fef588a7e5f"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-25 10:01:09.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca24b7a5e39915463eb475c6daaa31857e89b78b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca24b7a5e39915463eb475c6daaa31857e89b78b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b032a4430eea027d7807610673888fef588a7e5f","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":34,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5956166d6a766a5924ab85a7bf7e4e7fdee684c1":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2015-12-25 18:52:10.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/41"}}},"commit":{"parents":[{"commit":"bd68f5ea5219d2de38ce15f8fc9fe6a346a0abdb","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd68f5ea5219d2de38ce15f8fc9fe6a346a0abdb"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-25 18:51:57.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5956166d6a766a5924ab85a7bf7e4e7fdee684c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5956166d6a766a5924ab85a7bf7e4e7fdee684c1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bd68f5ea5219d2de38ce15f8fc9fe6a346a0abdb","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":35,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0ec69e8e29f8e399f9aeb74e4c2af48b89bfec78":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2015-12-26 09:46:22.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/42"}}},"commit":{"parents":[{"commit":"76e23491da5097fcfe29290551e308b151e216a4","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/76e23491da5097fcfe29290551e308b151e216a4"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-26 09:45:59.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ec69e8e29f8e399f9aeb74e4c2af48b89bfec78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ec69e8e29f8e399f9aeb74e4c2af48b89bfec78"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"76e23491da5097fcfe29290551e308b151e216a4","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":36,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e52b01fa206431656fdcd636c48eae9038e5eed4":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2015-12-26 10:38:10.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/43"}}},"commit":{"parents":[{"commit":"824fed4297b43f32e8ae2ca1b2fd479fddaf1256","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/824fed4297b43f32e8ae2ca1b2fd479fddaf1256"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-26 10:37:43.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e52b01fa206431656fdcd636c48eae9038e5eed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e52b01fa206431656fdcd636c48eae9038e5eed4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"824fed4297b43f32e8ae2ca1b2fd479fddaf1256","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":37,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0d49aea00946f3489cf033337c78981140ccc8e1":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2015-12-26 23:03:33.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/44"}}},"commit":{"parents":[{"commit":"8ad74b8ae5e230aa74adebcc32f64c965a8ee197","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ad74b8ae5e230aa74adebcc32f64c965a8ee197"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-26 23:03:12.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d49aea00946f3489cf033337c78981140ccc8e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d49aea00946f3489cf033337c78981140ccc8e1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8ad74b8ae5e230aa74adebcc32f64c965a8ee197","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":38,"change_status":"MERGED"}],"branch":"refs/heads/master"},"63e9c014529c8f9776ad3e9ae78f646c203ce875":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2015-12-27 14:30:31.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/45"}}},"commit":{"parents":[{"commit":"38026e90bb4c37c998aa166c4d4a7112f7577b4f","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38026e90bb4c37c998aa166c4d4a7112f7577b4f"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-27 14:26:16.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63e9c014529c8f9776ad3e9ae78f646c203ce875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63e9c014529c8f9776ad3e9ae78f646c203ce875"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38026e90bb4c37c998aa166c4d4a7112f7577b4f","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":39,"change_status":"MERGED"}],"branch":"refs/heads/master"},"07fdc8d0d9526291d8430fb2ea214ab111aa648d":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2015-12-27 19:48:42.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/46"}}},"commit":{"parents":[{"commit":"6c83222f62a97007b3e892859e6b900916ecbfae","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c83222f62a97007b3e892859e6b900916ecbfae"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-27 19:09:31.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07fdc8d0d9526291d8430fb2ea214ab111aa648d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07fdc8d0d9526291d8430fb2ea214ab111aa648d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c83222f62a97007b3e892859e6b900916ecbfae","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":40,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3de2b401b1ba16303d511f39deb0bf9745917146":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2015-12-27 22:23:53.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/47"}}},"commit":{"parents":[{"commit":"0772e99490d060a5e971408bf04099ee41c1fb1b","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0772e99490d060a5e971408bf04099ee41c1fb1b"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-27 22:17:19.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3de2b401b1ba16303d511f39deb0bf9745917146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3de2b401b1ba16303d511f39deb0bf9745917146"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0772e99490d060a5e971408bf04099ee41c1fb1b","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":41,"change_status":"MERGED"}],"branch":"refs/heads/master"},"874b98bf06b4e102e01aa4fb6bbab3f50e667f3e":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2015-12-28 15:43:22.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/48"}}},"commit":{"parents":[{"commit":"eeb362a04e81b28a754ed7219323394b3f3e0bff","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eeb362a04e81b28a754ed7219323394b3f3e0bff"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-28 15:42:17.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/874b98bf06b4e102e01aa4fb6bbab3f50e667f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/874b98bf06b4e102e01aa4fb6bbab3f50e667f3e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eeb362a04e81b28a754ed7219323394b3f3e0bff","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":42,"change_status":"MERGED"}],"branch":"refs/heads/master"},"051ad26193a322a0183cd42cd0359649f7b7b564":{"kind":"REWORK","_number":49,"created":"2015-12-28 17:08:39.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/49"}}},"commit":{"parents":[{"commit":"eeb362a04e81b28a754ed7219323394b3f3e0bff","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eeb362a04e81b28a754ed7219323394b3f3e0bff"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-28 17:07:38.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/051ad26193a322a0183cd42cd0359649f7b7b564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/051ad26193a322a0183cd42cd0359649f7b7b564"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eeb362a04e81b28a754ed7219323394b3f3e0bff","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":42,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2afc8d4f685afc3e9db65f6a5e8ca97f042ba58a":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2015-12-29 13:55:27.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/50"}}},"commit":{"parents":[{"commit":"dc148db10603c91436037e9a8a507b68d1bbac3a","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc148db10603c91436037e9a8a507b68d1bbac3a"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-29 13:54:50.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2afc8d4f685afc3e9db65f6a5e8ca97f042ba58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2afc8d4f685afc3e9db65f6a5e8ca97f042ba58a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dc148db10603c91436037e9a8a507b68d1bbac3a","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":43,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c6925e8d78e53d7b3ce960bd81206b9221150d06":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2015-12-30 07:59:27.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/51"}}},"commit":{"parents":[{"commit":"f8736aba1b2b5ace99808b1bc7e0e9fbcf4d0ba3","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8736aba1b2b5ace99808b1bc7e0e9fbcf4d0ba3"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2015-12-30 07:59:13.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6925e8d78e53d7b3ce960bd81206b9221150d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6925e8d78e53d7b3ce960bd81206b9221150d06"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f8736aba1b2b5ace99808b1bc7e0e9fbcf4d0ba3","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":44,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e53859fb83ff0b64f3a922326e7c2702c2de43bc":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2016-01-04 07:49:54.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/52"}}},"commit":{"parents":[{"commit":"2177220980edc006480a8a379755d8640cdcaf46","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2177220980edc006480a8a379755d8640cdcaf46"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 07:49:35.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e53859fb83ff0b64f3a922326e7c2702c2de43bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e53859fb83ff0b64f3a922326e7c2702c2de43bc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2177220980edc006480a8a379755d8640cdcaf46","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":45,"change_status":"MERGED"}],"branch":"refs/heads/master"},"92ecf4c647f2661ec0b9c450cdc37bebdb0e92fd":{"kind":"REWORK","_number":53,"created":"2016-01-04 08:07:37.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/53"}}},"commit":{"parents":[{"commit":"63357568466f95a843d9274282d3353e254b8272","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63357568466f95a843d9274282d3353e254b8272"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 08:07:03.000000000","tz":60},"subject":"virt: Allow multiple volume attachments","message":"virt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/92ecf4c647f2661ec0b9c450cdc37bebdb0e92fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/92ecf4c647f2661ec0b9c450cdc37bebdb0e92fd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"63357568466f95a843d9274282d3353e254b8272","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":46,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c45887998bc3809e054f064133d27615f112f05f":{"kind":"REWORK","_number":54,"created":"2016-01-04 15:35:03.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/54"}}},"commit":{"parents":[{"commit":"07f4c93ee94a07a2e0487313cfd59179fd8d3fcd","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07f4c93ee94a07a2e0487313cfd59179fd8d3fcd"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 14:33:05.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c45887998bc3809e054f064133d27615f112f05f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c45887998bc3809e054f064133d27615f112f05f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"07f4c93ee94a07a2e0487313cfd59179fd8d3fcd","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":47,"change_status":"MERGED"}],"branch":"refs/heads/master"},"aaf8c85d4734e22a39d48197f513679bd72966e2":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2016-01-04 18:12:44.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/55"}}},"commit":{"parents":[{"commit":"d402bfd8c84d4979109445695418516bdc55d6ff","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d402bfd8c84d4979109445695418516bdc55d6ff"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 18:12:25.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aaf8c85d4734e22a39d48197f513679bd72966e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aaf8c85d4734e22a39d48197f513679bd72966e2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d402bfd8c84d4979109445695418516bdc55d6ff","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":48,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f16fad02afe148428bd74d35a7689d5a42313b92":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2016-01-04 18:24:28.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/56"}}},"commit":{"parents":[{"commit":"49846c06eca9acfe1072bdd47d20305e5507085c","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/49846c06eca9acfe1072bdd47d20305e5507085c"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 18:24:13.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f16fad02afe148428bd74d35a7689d5a42313b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f16fad02afe148428bd74d35a7689d5a42313b92"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"49846c06eca9acfe1072bdd47d20305e5507085c","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":49,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e043a750809ffac6cb1d985a5844dc4c417bfbf5":{"kind":"REWORK","_number":57,"created":"2016-01-04 21:47:42.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/57"}}},"commit":{"parents":[{"commit":"49846c06eca9acfe1072bdd47d20305e5507085c","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/49846c06eca9acfe1072bdd47d20305e5507085c"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 21:27:27.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e043a750809ffac6cb1d985a5844dc4c417bfbf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e043a750809ffac6cb1d985a5844dc4c417bfbf5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"49846c06eca9acfe1072bdd47d20305e5507085c","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":49,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9ce678c5ae862834c146bb9d32f25423796fee8f":{"kind":"REWORK","_number":58,"created":"2016-01-04 22:20:41.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/58"}}},"commit":{"parents":[{"commit":"49846c06eca9acfe1072bdd47d20305e5507085c","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/49846c06eca9acfe1072bdd47d20305e5507085c"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-04 22:03:35.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ce678c5ae862834c146bb9d32f25423796fee8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ce678c5ae862834c146bb9d32f25423796fee8f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"49846c06eca9acfe1072bdd47d20305e5507085c","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":49,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e13e18842902aac80822213adc2521f9d773bc24":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2016-01-05 14:31:46.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/59"}}},"commit":{"parents":[{"commit":"e424bdea0de80039abc377a7a6e11f509e9b9650","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e424bdea0de80039abc377a7a6e11f509e9b9650"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-05 14:31:23.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e13e18842902aac80822213adc2521f9d773bc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e13e18842902aac80822213adc2521f9d773bc24"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e424bdea0de80039abc377a7a6e11f509e9b9650","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":50,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bd1dbc70d6febf63db43180d7e196d881f39ae01":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2016-01-05 23:22:05.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/60"}}},"commit":{"parents":[{"commit":"e00cb80979a414ba2198ff59ba378f7840569e75","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e00cb80979a414ba2198ff59ba378f7840569e75"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-05 23:18:30.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd1dbc70d6febf63db43180d7e196d881f39ae01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd1dbc70d6febf63db43180d7e196d881f39ae01"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e00cb80979a414ba2198ff59ba378f7840569e75","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":51,"change_status":"MERGED"}],"branch":"refs/heads/master"},"53467da3ed38d92d1dffc141a5f0491011c3b2fa":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2016-01-05 23:25:06.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/61"}}},"commit":{"parents":[{"commit":"12fac751e86e99f38344a5b15466cb20b175066e","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12fac751e86e99f38344a5b15466cb20b175066e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-05 23:24:34.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/53467da3ed38d92d1dffc141a5f0491011c3b2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/53467da3ed38d92d1dffc141a5f0491011c3b2fa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12fac751e86e99f38344a5b15466cb20b175066e","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":52,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b3b04d51bb34c578ead5d7365ce5c6e2566a6879":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2016-01-06 00:08:22.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/63"}}},"commit":{"parents":[{"commit":"a26fcb346fdf9ae6601862c6e2e2e14c86c963d7","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a26fcb346fdf9ae6601862c6e2e2e14c86c963d7"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-06 00:08:07.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3b04d51bb34c578ead5d7365ce5c6e2566a6879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3b04d51bb34c578ead5d7365ce5c6e2566a6879"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a26fcb346fdf9ae6601862c6e2e2e14c86c963d7","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":54,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1e628753326b748517f0b1acf2b7dce62e3caf4b":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2016-01-06 00:10:16.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/64"}}},"commit":{"parents":[{"commit":"0ab3a0bea9e60891e4bcd671f784b6b40aec3fd3","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ab3a0bea9e60891e4bcd671f784b6b40aec3fd3"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-06 00:10:00.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e628753326b748517f0b1acf2b7dce62e3caf4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e628753326b748517f0b1acf2b7dce62e3caf4b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0ab3a0bea9e60891e4bcd671f784b6b40aec3fd3","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":55,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ba3a260aa2249340aa2bc13a0e61dad9bd50872a":{"kind":"TRIVIAL_REBASE","_number":65,"created":"2016-01-06 00:13:28.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/65"}}},"commit":{"parents":[{"commit":"dedbe6b07d4168dace14904bbcf865e5ad361f4e","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dedbe6b07d4168dace14904bbcf865e5ad361f4e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-06 00:12:52.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba3a260aa2249340aa2bc13a0e61dad9bd50872a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba3a260aa2249340aa2bc13a0e61dad9bd50872a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dedbe6b07d4168dace14904bbcf865e5ad361f4e","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":56,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4ad0e480357cfc521a4f0b9d74d313b9ab939431":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2016-01-06 07:58:33.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/66"}}},"commit":{"parents":[{"commit":"dc8a641a49a6ebd84b010f100613e31ced12144c","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc8a641a49a6ebd84b010f100613e31ced12144c"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-06 07:58:11.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ad0e480357cfc521a4f0b9d74d313b9ab939431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ad0e480357cfc521a4f0b9d74d313b9ab939431"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dc8a641a49a6ebd84b010f100613e31ced12144c","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":57,"change_status":"MERGED"}],"branch":"refs/heads/master"},"16698d0c54a6272ec6073800ff6f43e2abe2e22c":{"kind":"REWORK","_number":67,"created":"2016-01-07 10:53:10.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/67"}}},"commit":{"parents":[{"commit":"f2a24858f64c9a5bafa44ad0393eee2123862eb5","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2a24858f64c9a5bafa44ad0393eee2123862eb5"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-07 09:01:50.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16698d0c54a6272ec6073800ff6f43e2abe2e22c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16698d0c54a6272ec6073800ff6f43e2abe2e22c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f2a24858f64c9a5bafa44ad0393eee2123862eb5","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":58,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8bf95cadc96ef148e1ae5f51cbaa0707dcc24053":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2016-01-07 17:53:07.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/68"}}},"commit":{"parents":[{"commit":"c95182b099aa2948390e858d6e2707b5b0e408d5","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c95182b099aa2948390e858d6e2707b5b0e408d5"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-07 17:52:36.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bf95cadc96ef148e1ae5f51cbaa0707dcc24053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bf95cadc96ef148e1ae5f51cbaa0707dcc24053"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c95182b099aa2948390e858d6e2707b5b0e408d5","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":59,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8242e720f30467b70ad433b5aaae219f6a8fa2e5":{"kind":"TRIVIAL_REBASE","_number":69,"created":"2016-01-07 17:57:02.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/69"}}},"commit":{"parents":[{"commit":"1f54503f7464398f7525a1d20ad36c5632a2079d","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f54503f7464398f7525a1d20ad36c5632a2079d"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-07 17:56:48.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8242e720f30467b70ad433b5aaae219f6a8fa2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8242e720f30467b70ad433b5aaae219f6a8fa2e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1f54503f7464398f7525a1d20ad36c5632a2079d","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":60,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8a2d54e800835ad691e6d4afaee9506826bb7783":{"kind":"REWORK","_number":70,"created":"2016-01-08 17:15:48.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/70"}}},"commit":{"parents":[{"commit":"05a8ac230cdcec6c2d4dc55a4c5edcd41bc23e27","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05a8ac230cdcec6c2d4dc55a4c5edcd41bc23e27"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-08 17:14:53.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a2d54e800835ad691e6d4afaee9506826bb7783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a2d54e800835ad691e6d4afaee9506826bb7783"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"05a8ac230cdcec6c2d4dc55a4c5edcd41bc23e27","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":61,"change_status":"MERGED"}],"branch":"refs/heads/master"},"78dc6918f6a337b17b9f583199886b7958bb9b14":{"kind":"TRIVIAL_REBASE","_number":71,"created":"2016-01-09 08:10:25.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/71"}}},"commit":{"parents":[{"commit":"05bc7f2b91a42814ab85403d480e497b2e41479e","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05bc7f2b91a42814ab85403d480e497b2e41479e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-09 08:07:38.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/78dc6918f6a337b17b9f583199886b7958bb9b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/78dc6918f6a337b17b9f583199886b7958bb9b14"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"05bc7f2b91a42814ab85403d480e497b2e41479e","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":62,"change_status":"MERGED"}],"branch":"refs/heads/master"},"382fc79a5a402b9da8befcf2210d74919af8ac7a":{"kind":"TRIVIAL_REBASE","_number":72,"created":"2016-01-12 10:05:36.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/72"}}},"commit":{"parents":[{"commit":"46683b7f8b55d509b1d65fa6ed7f1b1ff903c053","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46683b7f8b55d509b1d65fa6ed7f1b1ff903c053"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-12 08:21:06.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/382fc79a5a402b9da8befcf2210d74919af8ac7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/382fc79a5a402b9da8befcf2210d74919af8ac7a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"46683b7f8b55d509b1d65fa6ed7f1b1ff903c053","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":63,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0eaafbc22ab57ec80e4c9ae93c2fb576d5cb42f8":{"kind":"TRIVIAL_REBASE","_number":73,"created":"2016-01-12 16:37:23.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/73"}}},"commit":{"parents":[{"commit":"0ef7d1b4d2a634e78b0ad36b4a02b568d1e51e55","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ef7d1b4d2a634e78b0ad36b4a02b568d1e51e55"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-12 15:52:54.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nFor more information on how Cinder handles the attachments see\nthe reference to the Cinder and the Cinder client patches below.\nAlso for further details on volume attachment on Nova side please\ncheck the blueprint.\n\nThe API patch is part of a follow up spec that captures only the\nAPI changes in Nova.\n\nPartially-implements: blueprint multi-attach-volume\n\nRelated patches:\n\npython-cinderclient:\nhttps://review.openstack.org/#/c/85856/\n\ncinder:\nhttps://review.openstack.org/#/c/85847/\n\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0eaafbc22ab57ec80e4c9ae93c2fb576d5cb42f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0eaafbc22ab57ec80e4c9ae93c2fb576d5cb42f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0ef7d1b4d2a634e78b0ad36b4a02b568d1e51e55","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":64,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e4f9f93842a51cdbe03e5ae2ba9016329cf8f892":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":74,"created":"2016-01-12 20:19:45.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/74"}}},"commit":{"parents":[{"commit":"2f539dc90bf630cee86cb84e3b6506da314d9efe","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f539dc90bf630cee86cb84e3b6506da314d9efe"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-12 20:18:01.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e4f9f93842a51cdbe03e5ae2ba9016329cf8f892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e4f9f93842a51cdbe03e5ae2ba9016329cf8f892"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2f539dc90bf630cee86cb84e3b6506da314d9efe","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":65,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dbd9712db95825c68c78fc7184ae2b127bee7922":{"kind":"REWORK","_number":75,"created":"2016-01-12 23:39:39.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/75"}}},"commit":{"parents":[{"commit":"ebb41e767872f9ad5287f897b4bb7581d0d4bfbb","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ebb41e767872f9ad5287f897b4bb7581d0d4bfbb"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-12 23:39:16.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dbd9712db95825c68c78fc7184ae2b127bee7922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dbd9712db95825c68c78fc7184ae2b127bee7922"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ebb41e767872f9ad5287f897b4bb7581d0d4bfbb","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":66,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2b2803e9562dd5aa108a87ecc412a28cabe01607":{"kind":"TRIVIAL_REBASE","_number":76,"created":"2016-01-13 06:51:10.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/76"}}},"commit":{"parents":[{"commit":"7aa1e5d6b0c632812c703defc59e02d5b6309c1c","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7aa1e5d6b0c632812c703defc59e02d5b6309c1c"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-13 06:50:48.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2b2803e9562dd5aa108a87ecc412a28cabe01607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2b2803e9562dd5aa108a87ecc412a28cabe01607"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7aa1e5d6b0c632812c703defc59e02d5b6309c1c","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":67,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ffbf483078366fb612daea2f730e9b6fbe99370b":{"kind":"REWORK","_number":77,"created":"2016-01-13 09:21:18.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/77"}}},"commit":{"parents":[{"commit":"c3fcc71a15b4fcdeafb9952fcfd75bb585a089ff","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3fcc71a15b4fcdeafb9952fcfd75bb585a089ff"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-13 09:20:54.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ffbf483078366fb612daea2f730e9b6fbe99370b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ffbf483078366fb612daea2f730e9b6fbe99370b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c3fcc71a15b4fcdeafb9952fcfd75bb585a089ff","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":68,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ec635f5b2183ba3acbca2d4da94f2422d8539ab8":{"kind":"REWORK","_number":78,"created":"2016-01-13 11:49:54.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/78"}}},"commit":{"parents":[{"commit":"dbd175983019eff57d0d7497e6a9264c819d5e22","subject":"comp: Allow multiple volume attachments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dbd175983019eff57d0d7497e6a9264c819d5e22"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-13 11:34:37.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec635f5b2183ba3acbca2d4da94f2422d8539ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec635f5b2183ba3acbca2d4da94f2422d8539ab8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dbd175983019eff57d0d7497e6a9264c819d5e22","is_merged_in_target_branch":false,"change_id":"I3cdc49924acbdd21d1e6678a3bb4cf7de7f1db1a","change_number":193134,"patch_set_number":69,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bb26b5ea108c219b7eab85caa34ad339ad8034d2":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2016-01-14 14:08:47.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/79"}}},"commit":{"parents":[{"commit":"8749b1d73a742bb2bcdd6cc0f312fddedcd2bcb7","subject":"comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8749b1d73a742bb2bcdd6cc0f312fddedcd2bcb7"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-14 14:05:21.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb26b5ea108c219b7eab85caa34ad339ad8034d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb26b5ea108c219b7eab85caa34ad339ad8034d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8749b1d73a742bb2bcdd6cc0f312fddedcd2bcb7","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0a156f5ac5ff3f105692a173c103c7178b1d3034":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":80,"created":"2016-01-15 13:29:15.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/80"}}},"commit":{"parents":[{"commit":"567a19aac683efe72cf93384e0b9579d5310f8ec","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/567a19aac683efe72cf93384e0b9579d5310f8ec"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-15 13:28:49.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a156f5ac5ff3f105692a173c103c7178b1d3034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a156f5ac5ff3f105692a173c103c7178b1d3034"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"567a19aac683efe72cf93384e0b9579d5310f8ec","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7e9f70f5667e951a375b5d249ae9091ae492f54a":{"kind":"TRIVIAL_REBASE","_number":81,"created":"2016-01-15 22:44:17.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/81"}}},"commit":{"parents":[{"commit":"5e4df85e64d3e810b320112a85c5182f66c5edcb","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e4df85e64d3e810b320112a85c5182f66c5edcb"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-15 22:43:30.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e9f70f5667e951a375b5d249ae9091ae492f54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e9f70f5667e951a375b5d249ae9091ae492f54a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e4df85e64d3e810b320112a85c5182f66c5edcb","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e29d8165f5c8ba67be043d940e602e67985ae99f":{"kind":"TRIVIAL_REBASE","_number":82,"created":"2016-01-16 10:09:18.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/82"}}},"commit":{"parents":[{"commit":"d8380ed250579dd030f738a6c7fa9932db7139ef","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8380ed250579dd030f738a6c7fa9932db7139ef"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-16 10:09:02.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e29d8165f5c8ba67be043d940e602e67985ae99f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e29d8165f5c8ba67be043d940e602e67985ae99f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d8380ed250579dd030f738a6c7fa9932db7139ef","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c620a9be613f6a069e1a73b02730e0b184e4c1f0":{"kind":"REWORK","_number":83,"created":"2016-01-16 17:12:49.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/83"}}},"commit":{"parents":[{"commit":"61d1c8171fcd9ed149619d5e9ea2faa0a968abbf","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61d1c8171fcd9ed149619d5e9ea2faa0a968abbf"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-16 17:12:22.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c620a9be613f6a069e1a73b02730e0b184e4c1f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c620a9be613f6a069e1a73b02730e0b184e4c1f0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"61d1c8171fcd9ed149619d5e9ea2faa0a968abbf","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"298e529654fc8f05ee3a5aeeec66e7baabaa0bae":{"kind":"TRIVIAL_REBASE","_number":84,"created":"2016-01-16 20:31:44.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/84"}}},"commit":{"parents":[{"commit":"4f2c557c070fb2ef7aad6025abc39a8146824953","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f2c557c070fb2ef7aad6025abc39a8146824953"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-16 20:20:49.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/298e529654fc8f05ee3a5aeeec66e7baabaa0bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/298e529654fc8f05ee3a5aeeec66e7baabaa0bae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4f2c557c070fb2ef7aad6025abc39a8146824953","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"86b0ba46ecc90ebc39a4f735d12cb7db36ef333e":{"kind":"TRIVIAL_REBASE","_number":85,"created":"2016-01-19 00:50:05.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/85"}}},"commit":{"parents":[{"commit":"66f0fdabc73cbba1e31e731df9e74c37a804ab5e","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66f0fdabc73cbba1e31e731df9e74c37a804ab5e"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-19 00:50:03.000000000","tz":0},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/86b0ba46ecc90ebc39a4f735d12cb7db36ef333e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/86b0ba46ecc90ebc39a4f735d12cb7db36ef333e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"66f0fdabc73cbba1e31e731df9e74c37a804ab5e","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a3e2c1e9b10b6c6b3a86437d7454cb6779a359b1":{"kind":"TRIVIAL_REBASE","_number":86,"created":"2016-01-19 05:48:44.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/86"}}},"commit":{"parents":[{"commit":"7d53109cc7532c564364d751553570f9a8b60ee3","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d53109cc7532c564364d751553570f9a8b60ee3"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-19 05:48:31.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3e2c1e9b10b6c6b3a86437d7454cb6779a359b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3e2c1e9b10b6c6b3a86437d7454cb6779a359b1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7d53109cc7532c564364d751553570f9a8b60ee3","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b8bcafd0d6e2901f1bb2de31616ffe0c599527cd":{"kind":"TRIVIAL_REBASE","_number":87,"created":"2016-01-19 18:41:00.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/87"}}},"commit":{"parents":[{"commit":"27b91f5083068cd3a73c133e529bad31dfb041a4","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/27b91f5083068cd3a73c133e529bad31dfb041a4"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-19 18:38:35.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8bcafd0d6e2901f1bb2de31616ffe0c599527cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8bcafd0d6e2901f1bb2de31616ffe0c599527cd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"27b91f5083068cd3a73c133e529bad31dfb041a4","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"07b49532de5afe223f0f725d6c204ea774045a6c":{"kind":"TRIVIAL_REBASE","_number":88,"created":"2016-01-19 18:46:43.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/88"}}},"commit":{"parents":[{"commit":"f50727a188f93f9ba40ebfd1507fb9f8e8386d82","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f50727a188f93f9ba40ebfd1507fb9f8e8386d82"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-19 18:46:18.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07b49532de5afe223f0f725d6c204ea774045a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07b49532de5afe223f0f725d6c204ea774045a6c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f50727a188f93f9ba40ebfd1507fb9f8e8386d82","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0699703d0dcb99d7fa609c0f5c1517f097a0f705":{"kind":"TRIVIAL_REBASE","_number":89,"created":"2016-01-19 20:47:53.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/89"}}},"commit":{"parents":[{"commit":"ef44b5323b3df27b93a6c75c159c8441705d90ca","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef44b5323b3df27b93a6c75c159c8441705d90ca"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-19 20:47:27.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0699703d0dcb99d7fa609c0f5c1517f097a0f705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0699703d0dcb99d7fa609c0f5c1517f097a0f705"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ef44b5323b3df27b93a6c75c159c8441705d90ca","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"46ad7f535817592f86fe573f69e95d89c0ab17f5":{"kind":"TRIVIAL_REBASE","_number":90,"created":"2016-01-20 18:09:01.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/90"}}},"commit":{"parents":[{"commit":"f6fc69a1d2a459583c1eb0e3cb790d32a2e3159d","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6fc69a1d2a459583c1eb0e3cb790d32a2e3159d"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-20 18:06:41.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46ad7f535817592f86fe573f69e95d89c0ab17f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46ad7f535817592f86fe573f69e95d89c0ab17f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f6fc69a1d2a459583c1eb0e3cb790d32a2e3159d","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4fd2ac323bb1d181f5a1255679da240464cbd207":{"kind":"TRIVIAL_REBASE","_number":91,"created":"2016-01-20 21:05:25.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/91"}}},"commit":{"parents":[{"commit":"1366d3ffb0ac931ef736e892eb0493bb268271ab","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1366d3ffb0ac931ef736e892eb0493bb268271ab"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-20 21:05:05.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4fd2ac323bb1d181f5a1255679da240464cbd207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4fd2ac323bb1d181f5a1255679da240464cbd207"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1366d3ffb0ac931ef736e892eb0493bb268271ab","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cc9b56aa74dc352119f6795b4ffd7f5e933635e3":{"kind":"TRIVIAL_REBASE","_number":92,"created":"2016-01-20 21:18:20.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/92"}}},"commit":{"parents":[{"commit":"ddffc4e83c8ba48818eaea32186622375bf3025d","subject":"WIP: comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ddffc4e83c8ba48818eaea32186622375bf3025d"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-20 21:17:00.000000000","tz":60},"subject":"WIP: libvirt: Allow multiple volume attachments","message":"WIP: libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc9b56aa74dc352119f6795b4ffd7f5e933635e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc9b56aa74dc352119f6795b4ffd7f5e933635e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ddffc4e83c8ba48818eaea32186622375bf3025d","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ca3f90a601ed88518bdb9ea5b5e6e141043ac87a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":93,"created":"2016-01-21 14:11:15.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/93"}}},"commit":{"parents":[{"commit":"c53ccf533dc97053c42458f741deee1d433fb152","subject":"comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c53ccf533dc97053c42458f741deee1d433fb152"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-21 12:01:29.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca3f90a601ed88518bdb9ea5b5e6e141043ac87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca3f90a601ed88518bdb9ea5b5e6e141043ac87a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c53ccf533dc97053c42458f741deee1d433fb152","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a62d37c106473e172a8cd80270b848843af8ffab":{"kind":"REWORK","_number":94,"created":"2016-01-21 18:17:52.000000000","uploader":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"ref":"refs/changes/35/193135/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/193135/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/193135/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/193135/94"}}},"commit":{"parents":[{"commit":"42fbe739a97cfd4758faaf8a79276e4692c95194","subject":"comp: Allow volume multiattach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/42fbe739a97cfd4758faaf8a79276e4692c95194"}]}],"author":{"name":"Krisztian Gacsal","email":"krisztian.gacsal@ericsson.com","date":"2015-06-18 12:05:21.000000000","tz":120},"committer":{"name":"Ildiko Vancsa","email":"ildiko.vancsa@ericsson.com","date":"2016-01-21 18:07:06.000000000","tz":60},"subject":"libvirt: Allow multiple volume attachments","message":"libvirt: Allow multiple volume attachments\n\nAdd multiattach support to libvirt driver, by updating the\nxml configuration info if the multiattach support is turned\non for a volume and set the virt driver capability\n\u0027support_multiattach\u0027 to true.\n\nAlso the attach function in nova/virt/block_device.py is updated\nto call out to Cinder in case of each attachment request for\nmultiattach volumes, which is need for Cinder in order to track\nall attachments for a volume and be able to detach properly.\n\nCo-Authored-By: Ildiko Vancsa \u003cildiko.vancsa@ericsson.com\u003e\n\nPartially-implements: blueprint multi-attach-volume\nChange-Id: Ib78fcd7c54c966dcbc143165a262e582bbda0901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a62d37c106473e172a8cd80270b848843af8ffab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a62d37c106473e172a8cd80270b848843af8ffab"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"42fbe739a97cfd4758faaf8a79276e4692c95194","is_merged_in_target_branch":false,"change_id":"I947bf0ad34a48e9182a3dc016f47f0c9f71c9d7b","change_number":267587,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
