)]}'
{"id":"openstack%2Fnova~197568","triplet_id":"openstack%2Fnova~master~Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62","project":"openstack/nova","branch":"master","topic":"bp/libvirt-real-time","hashtags":[],"change_id":"Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62","subject":"virt: add constraint to handle realtime policy","status":"MERGED","created":"2015-07-01 13:03:51.000000000","updated":"2015-11-19 14:31:35.000000000","submitted":"2015-11-19 14:31:31.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"333cbdaf0f6f253a8f4587451f1370a7bdad00bf","_number":197568,"virtual_id_number":197568,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2015-11-06 10:53:09.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-11-06 10:57:25.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},{"value":2,"date":"2015-11-19 14:31:31.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-11-04 14:17:33.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2015-11-03 06:41:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2015-11-03 06:58:35.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"date":"2015-11-03 06:16:36.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2015-11-03 06:11:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":1,"date":"2015-11-05 18:24:57.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2015-11-03 06:43:09.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-11-06 10:39:22.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2015-11-06 09:02:23.000000000","_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":1,"date":"2015-11-03 07:07:02.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"date":"2015-11-03 07:33:30.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"recommended":{"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},"disliked":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-11-03 05:44:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":-1,"date":"2015-11-03 05:44:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":2,"date":"2015-11-03 05:44:31.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-11-03 10:41:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2015-11-19 12:00:41.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-11-19 12:00:41.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-10-05 11:09:19.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2015-10-07 12:43:08.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2015-10-08 09:30:11.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2015-10-09 15:06:45.000000000","updated_by":{"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},"reviewer":{"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},"state":"REVIEWER"},{"updated":"2015-11-03 06:11:32.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-03 06:16:36.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-03 06:41:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-03 06:43:09.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-03 06:58:35.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-11-03 07:07:02.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-03 07:33:30.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-03 10:41:52.000000000","updated_by":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"reviewer":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"state":"REVIEWER"},{"updated":"2015-11-04 14:17:33.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-05 18:24:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-06 10:39:22.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-06 10:53:09.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-06 10:57:25.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-19 12:00:41.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2015-11-19 14:31:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"98dd12353300c6876953f17a8e59d20357b79e0b","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-07-01 13:03:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2683da434d35a74901c889b7f378e2e56f3a3d0b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-01 13:50:35.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/ef24e39ebd8942ce8d05898635c0298c/index.html : SUCCESS in 20m 21s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/03ea7711c85f4be784ea86bf4d2600c0/index.html : SUCCESS in 25m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f306ee1152f64f1bbc917dd19f037832/index.html : SUCCESS in 21m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/1/check/real-db-upgrade_nova_percona_user_001:th-percona/50e1c28ff71f4d469d27b69df8b540f5/index.html : SUCCESS in 27m 52s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/23cdb3c2d36849e790154e73ea1daf8e/index.html : SUCCESS in 21m 55s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/1/check/real-db-upgrade_nova_percona_user_002:th-percona/22e261b5ed0e42718700b21d98b3e763/index.html : SUCCESS in 21m 07s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"7788c3a9b9d0a155fa6a3923e35c8c76b9a1a400","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-07-01 13:53:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e410a5b1003c7e19d4777504c218693be14de77c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-01 15:07:00.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/2/2951/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"b463eba3e991453d7a2cf99987a68df0baec3834","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-01 15:13:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/2/check/dsvm-tempest-xen/2d89534 : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b3f01867f9f03ce34f2a4f2ca1d72b471fe12c17","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-01 15:26:52.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b659cb7ba4f3446587ad08b6f394d488/index.html : SUCCESS in 21m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5557e467aa6e4c7395b094575d2c1653/index.html : SUCCESS in 25m 51s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/56fabb1ed754467dbccfb46fe0700e4e/index.html : SUCCESS in 17m 34s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/2/check/real-db-upgrade_nova_percona_user_001:th-percona/3b19769cf65d4077baeb2f37ac80b289/index.html : SUCCESS in 29m 00s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/cb1453051c344094880e0c7104711839/index.html : SUCCESS in 20m 45s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/2/check/real-db-upgrade_nova_percona_user_002:th-percona/a33150a105414517a9c5e10a1614ebe8/index.html : SUCCESS in 22m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"93f5777ec8f1ac93b60f60c4d4823ad7136b5334","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-01 15:31:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/2/check/gate-nova-pep8/23231b9/ : SUCCESS in 6m 43s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/2/check/gate-nova-docs/4b0a683//doc/build/html/ : SUCCESS in 6m 54s\n- gate-nova-python27 http://logs.openstack.org/68/197568/2/check/gate-nova-python27/e4c406a/ : SUCCESS in 13m 17s\n- gate-nova-python34 http://logs.openstack.org/68/197568/2/check/gate-nova-python34/5fc5af5/ : SUCCESS in 8m 52s\n- check-tempest-dsvm-full http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-full/22da8f5/ : SUCCESS in 34m 41s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-postgres-full/8e569f4/ : SUCCESS in 42m 18s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-neutron-full/7044295/ : SUCCESS in 58m 01s\n- check-grenade-dsvm http://logs.openstack.org/68/197568/2/check/check-grenade-dsvm/48ecb6b/ : SUCCESS in 39m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/2/check/gate-tempest-dsvm-large-ops/5932697/ : SUCCESS in 28m 02s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/2/check/gate-tempest-dsvm-neutron-large-ops/fdb8f3d/ : SUCCESS in 33m 47s\n- check-devstack-dsvm-cells http://logs.openstack.org/68/197568/2/check/check-devstack-dsvm-cells/db66f3c/ : SUCCESS in 25m 05s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/2/check/gate-nova-tox-functional/971dd6e/ : SUCCESS in 12m 25s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/2/check/check-grenade-dsvm-partial-ncpu/3fa9534/ : SUCCESS in 41m 22s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-ironic-pxe_ssh/526f886/ : SUCCESS in 31m 50s\n- check-tempest-dsvm-docker http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-docker/8be68a6/ : SUCCESS in 34m 40s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-nova-v21-full/3d9de6a/ : SUCCESS in 45m 57s\n- check-tempest-dsvm-cells http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-cells/c368eb5/ : SUCCESS in 49m 36s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-full-ceph/675fd09/ : SUCCESS in 37m 51s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-multinode-full/d390bec/ : FAILURE in 1h 20m 21s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-multinode-smoke/a61156e/ : FAILURE in 18m 55s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/68/197568/2/check/check-tempest-dsvm-neutron-multinode-smoke/df8d520/ : FAILURE in 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42c3bca20d329919329a1fa366ede263831f9442","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-01 17:13:12.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/2/check-ibm-tempest-dsvm-full/d1d793f : SUCCESS in 46m 49s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/2/check-ibm-tempest-dsvm-postgres-full/345140f : SUCCESS in 48m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4da8b7d2b26127f57e3e958dca03674cca70c2dc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-01 18:13:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/197568/2/ : SUCCESS in 1h 45m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2c25fc07201745a9b955021fc11632acfa6a141f","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-07-09 10:03:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"870692a844e20da6c58d892afaca3f9355e54618","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-09 10:54:34.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/49b1b05eba764be0bb461113e2158dd0/index.html : SUCCESS in 20m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d4b042a0ef304331aa6df6327c4667a5/index.html : SUCCESS in 28m 30s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/165b9ad626d0482f907abca56af9943b/index.html : SUCCESS in 18m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/3/check/real-db-upgrade_nova_percona_user_001:th-percona/733283cb2a0b4da8a9f38cd0b1e6b4c2/index.html : SUCCESS in 23m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/136c97a00e5544e7aaee767f23c22f38/index.html : SUCCESS in 21m 28s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/3/check/real-db-upgrade_nova_percona_user_002:th-percona/3063e6fcbd494d0eb508a9ffaa065aac/index.html : SUCCESS in 22m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"796858fdce25eb6b281b872a368ac37b349251aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-09 11:06:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/3/check/gate-nova-pep8/5522561/ : SUCCESS in 8m 58s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/3/check/gate-nova-docs/368aa55//doc/build/html/ : SUCCESS in 9m 34s\n- gate-nova-python27 http://logs.openstack.org/68/197568/3/check/gate-nova-python27/47819f5/ : SUCCESS in 14m 31s\n- gate-nova-python34 http://logs.openstack.org/68/197568/3/check/gate-nova-python34/89038b6/ : SUCCESS in 9m 30s\n- check-tempest-dsvm-full http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-full/293e203/ : SUCCESS in 42m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-postgres-full/2272b4b/ : SUCCESS in 48m 22s\n- check-tempest-dsvm-neutron http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-neutron/3fa433d/ : SUCCESS in 32m 02s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-neutron-full/f5778ef/ : FAILURE in 1h 00m 03s\n- check-grenade-dsvm http://logs.openstack.org/68/197568/3/check/check-grenade-dsvm/8dc9c50/ : SUCCESS in 52m 00s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/3/check/gate-tempest-dsvm-large-ops/9c70025/ : SUCCESS in 36m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/3/check/gate-tempest-dsvm-neutron-large-ops/40cfb8f/ : SUCCESS in 23m 47s\n- check-devstack-dsvm-cells http://logs.openstack.org/68/197568/3/check/check-devstack-dsvm-cells/a9bb091/ : SUCCESS in 28m 55s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/3/check/gate-nova-tox-functional/b8a7226/ : SUCCESS in 9m 54s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/3/check/check-grenade-dsvm-partial-ncpu/dcd77fb/ : SUCCESS in 43m 12s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-ironic-pxe_ssh/9ad238b/ : SUCCESS in 39m 36s\n- check-tempest-dsvm-docker http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-docker/e047ea7/ : SUCCESS in 45m 25s (non-voting)\n- check-tempest-dsvm-nova-v21-full http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-nova-v21-full/35db16a/ : SUCCESS in 50m 56s\n- check-tempest-dsvm-cells http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-cells/343cb2d/ : SUCCESS in 46m 35s\n- check-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-full-ceph/d9a031a/ : SUCCESS in 50m 52s\n- check-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-multinode-full/1c2067c/ : FAILURE in 24m 15s (non-voting)\n- check-tempest-dsvm-multinode-smoke http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-multinode-smoke/478dda1/ : FAILURE in 50m 24s (non-voting)\n- check-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/68/197568/3/check/check-tempest-dsvm-neutron-multinode-smoke/b328e38/ : FAILURE in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c1928aa43a986068f1ca2a7fd14551c1bef10f7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-07-09 11:11:54.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1313168\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"29815d0d3528d80988b1b470f25b0c78999d0298","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-09 11:14:00.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/3/check-ibm-tempest-dsvm-full/a083818 : SUCCESS in 50m 12s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/3/check-ibm-tempest-dsvm-postgres-full/995b2f0 : SUCCESS in 1h 08m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"623700b4e75feeef1f1ebeffb2fb61ade0bb2739","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-09 11:31:38.000000000","message":"Patch Set 3: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/3/3810/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"927c0fd3ee557bca11acb4427bbdf676fdf75141","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-09 12:09:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/3/check/dsvm-tempest-xen/525e7bf : SUCCESS in 1h 27m 18s","accounts_in_message":[],"_revision_number":3},{"id":"baee45657750c718c7a35ad6c56c8ad498f5a719","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-09 13:46:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/197568/3/ : SUCCESS in 1h 51m 03s","accounts_in_message":[],"_revision_number":3},{"id":"963a280110cee0e60d0415accba086e1ad201b5a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-10 05:12:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/2 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ec2aeae65802b1cf3126cd024a878ca6e77478fb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-14 22:53:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/3 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d772015e6e341f8200089ffab3ecb74128abe38b","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-07-15 12:50:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c06dbed0d41a9fa21e966a245baf195926105a41","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-15 13:11:24.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d4240ba8251843c2b428a80670b3a556/index.html : SUCCESS in 13m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b64f3f80123a4632926a91563c529f2a/index.html : SUCCESS in 14m 59s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/baf1a81d4dd64c2b8f5b5cf4428c5f7c/index.html : SUCCESS in 13m 48s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/4/check/real-db-upgrade_nova_percona_user_001:th-percona/a65d650557e94edaa0c985615836048b/index.html : SUCCESS in 14m 54s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/261f66b9ac8c407aba29e299e9a67a86/index.html : SUCCESS in 11m 51s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/4/check/real-db-upgrade_nova_percona_user_002:th-percona/9bc32f51cae44b73ae3d0b59175bf687/index.html : SUCCESS in 14m 07s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"103e92d65ccc97d814f666b81494da3b75c888b6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-15 13:47:25.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/4/check-ibm-tempest-dsvm-full/cf5b247 : SUCCESS in 55m 08s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/4/check-ibm-tempest-dsvm-postgres-full/e1c48fa : SUCCESS in 51m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c16732677b050b9265ed6e0a006d960d5c5ae3d2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-15 13:52:28.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/4/4329/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"ed2eda1c2ecf62dd5d57f526730b746ffef12d10","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-15 15:00:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/4/check/dsvm-tempest-xen/63bc8c2 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":4},{"id":"86618e4be9d9121ffe3e1a0e3c6e7029bc244267","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-15 15:00:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/4/check/gate-nova-pep8/91efdb8/ : SUCCESS in 13m 39s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/4/check/gate-nova-docs/0e57d69//doc/build/html/ : SUCCESS in 14m 04s\n- gate-nova-python27 http://logs.openstack.org/68/197568/4/check/gate-nova-python27/5f1dcee/ : SUCCESS in 17m 39s\n- gate-nova-python34 http://logs.openstack.org/68/197568/4/check/gate-nova-python34/94dca07/ : SUCCESS in 17m 58s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-full/e6ba30e/ : SUCCESS in 50m 41s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-postgres-full/6fb8a05/ : SUCCESS in 52m 51s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-neutron/2963422/ : SUCCESS in 44m 33s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-neutron-full/42a8c6f/ : SUCCESS in 1h 00m 20s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/4/check/gate-grenade-dsvm/a5f883a/ : SUCCESS in 49m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-large-ops/b70c740/ : SUCCESS in 29m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-neutron-large-ops/25b0ac4/ : SUCCESS in 27m 47s\n- gate-devstack-dsvm-cells http://logs.openstack.org/68/197568/4/check/gate-devstack-dsvm-cells/578d6e4/ : SUCCESS in 30m 22s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/4/check/gate-nova-tox-functional/660b125/ : SUCCESS in 16m 56s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/4/check/gate-grenade-dsvm-partial-ncpu/3068043/ : SUCCESS in 52m 24s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-ironic-pxe_ssh/6c36a85/ : SUCCESS in 37m 12s\n- gate-tempest-dsvm-docker http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-docker/6538bab/ : FAILURE in 9m 01s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-nova-v21-full/3857ff9/ : SUCCESS in 57m 05s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-cells/f4400f5/ : SUCCESS in 48m 09s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-full-ceph/73af51c/ : SUCCESS in 51m 59s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-multinode-full/3149841/ : NOT_REGISTERED (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-multinode-smoke/4a83d0b/ : NOT_REGISTERED (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/68/197568/4/check/gate-tempest-dsvm-neutron-multinode-smoke/5fdf8c4/ : NOT_REGISTERED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"65e4cc78047e3a5493aea1232fbe045a10fa7075","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-07-21 15:22:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"10d3b98cc17ebdc36a1173a00fcaa6c21344d97b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-21 16:28:26.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/720f737819fa4b57babb104c515b1a63/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e8a0060f7c404efcbb45f666eeca01c3/index.html : SUCCESS in 13m 15s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/49b5965980b546deaf8aeddc7b5c89bb/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/5/check/real-db-upgrade_nova_percona_user_001:th-percona/80e4e3b296824bcd807f099304ab98d1/index.html : SUCCESS in 16m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/bc83bb5244d24394a288ff996d49fb9d/index.html : SUCCESS in 12m 00s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/5/check/real-db-upgrade_nova_percona_user_002:th-percona/e2d8fde306e64f278be37503094c77cb/index.html : SUCCESS in 14m 23s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"2119199cb4e008a259293d4ef146820ee022b639","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-21 16:50:14.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/5/5112/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"c6622e3814d4d944977287f87c6e8ce33ebf2b11","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-21 16:59:14.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/5/check-ibm-tempest-dsvm-full/21c23da : SUCCESS in 1h 16m 48s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/5/check-ibm-tempest-dsvm-postgres-full/d6541e1 : SUCCESS in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"72856fb762a354ea93e73f2032fedbcc60d1ca58","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-21 17:47:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/5/check/dsvm-tempest-xen/adbc7ef : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":5},{"id":"93df3677c8d9bc4ac633a0914e053590389fc7fd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-21 18:11:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/5/check/gate-nova-pep8/9774c0c/ : SUCCESS in 7m 00s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/5/check/gate-nova-docs/afdb130//doc/build/html/ : SUCCESS in 8m 34s\n- gate-nova-python27 http://logs.openstack.org/68/197568/5/check/gate-nova-python27/24732bd/ : SUCCESS in 14m 31s\n- gate-nova-python34 http://logs.openstack.org/68/197568/5/check/gate-nova-python34/5d536c1/ : SUCCESS in 13m 40s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-full/6a0d8e0/ : SUCCESS in 59m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-postgres-full/97d50dc/ : SUCCESS in 1h 00m 18s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-neutron-full/1eb9c3a/ : FAILURE in 30m 22s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/5/check/gate-grenade-dsvm/a6da542/ : FAILURE in 26m 09s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-large-ops/d6394ea/ : SUCCESS in 27m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-neutron-large-ops/bb797a6/ : FAILURE in 58m 58s\n- gate-devstack-dsvm-cells http://logs.openstack.org/68/197568/5/check/gate-devstack-dsvm-cells/87d0bad/ : SUCCESS in 35m 37s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/5/check/gate-nova-tox-functional/5621daf/ : SUCCESS in 10m 21s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/5/check/gate-grenade-dsvm-partial-ncpu/530b216/ : FAILURE in 31m 54s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-ironic-pxe_ssh/876871f/ : SUCCESS in 42m 25s\n- gate-tempest-dsvm-docker http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-docker/16d12cd/ : FAILURE in 41m 09s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-nova-v21-full/7add0b1/ : SUCCESS in 54m 15s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-cells/103507c/ : SUCCESS in 47m 29s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-full-ceph/738dd77/ : SUCCESS in 40m 00s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-multinode-full/140de36/ : FAILURE in 1h 09m 13s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-multinode-smoke/124f0c2/ : SUCCESS in 43m 42s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/68/197568/5/check/gate-tempest-dsvm-neutron-multinode-smoke/db9cfe8/ : SUCCESS in 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"69793652e5264202a6c9aad7f9097d789d032f82","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-21 19:00:51.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/197568/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"6a07165acc08eab66dc004752be65f42021c374e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-21 22:52:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/5/Nova-ML2-Sriov : SUCCESS in 1h 14m 49s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"bb8bfac8d6f1add37437a909846600f6abaaebdc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-22 03:22:32.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/197568/5/ : FAILURE in 1h 48m 06s","accounts_in_message":[],"_revision_number":5},{"id":"6217d4df8d3079ae997d9ebeb283aec2c34d38c0","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-07-23 10:47:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"032c137055b9c1513cb4f90df2b83a003eb591a5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-23 11:10:06.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3f84bdda4b0e4d29beccb201dcf7902a/index.html : SUCCESS in 14m 11s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/215b9fc936324307bb26df6ec42e2c7f/index.html : SUCCESS in 15m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/97b9abb149bd4449902968686a162f49/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/6/check/real-db-upgrade_nova_percona_user_001:th-percona/33500d2d5b0345708d38fa9b997b6743/index.html : SUCCESS in 13m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3bf31e76743c4def9abb79155f19be79/index.html : SUCCESS in 12m 22s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/6/check/real-db-upgrade_nova_percona_user_002:th-percona/b8c822ec33314454a73d520e29ba4d8f/index.html : SUCCESS in 12m 12s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"119b859aaa4ce58368cc58f8d27355c8ecd5b5d1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-07-23 11:25:06.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/197568/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"a11d4632d83cc7bb4ffb76a52f92a5e10760db81","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-23 11:37:10.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/6/check-ibm-tempest-dsvm-full/c091199 : SUCCESS in 48m 17s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/6/check-ibm-tempest-dsvm-postgres-full/5af2ce8 : SUCCESS in 45m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3394ce8b3135641b64cf414aecaa3c2fb2f040f7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-23 12:03:38.000000000","message":"Patch Set 6: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/6/5409/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"1d0e5925b78f2daf966f48ce8bc26e6e884e1231","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-23 12:06:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/6/check/dsvm-tempest-xen/5f762b5 : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":6},{"id":"fd45a9f498b30efe684d4e061567173e7edd123b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 12:59:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/6/check/gate-nova-pep8/2f9bff2/ : SUCCESS in 10m 02s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/6/check/gate-nova-docs/c2ac902//doc/build/html/ : SUCCESS in 8m 42s\n- gate-nova-python27 http://logs.openstack.org/68/197568/6/check/gate-nova-python27/b6c07f0/ : SUCCESS in 14m 55s\n- gate-nova-python34 http://logs.openstack.org/68/197568/6/check/gate-nova-python34/acce2dc/ : SUCCESS in 13m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-full/a1c96f8/ : SUCCESS in 59m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-postgres-full/a3abf8e/ : SUCCESS in 1h 08m 43s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-neutron-full/bc05cb5/ : SUCCESS in 1h 19m 21s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/6/check/gate-grenade-dsvm/fa80023/ : SUCCESS in 54m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-large-ops/5de0ce4/ : SUCCESS in 42m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-neutron-large-ops/ed5270f/ : SUCCESS in 45m 53s\n- gate-devstack-dsvm-cells http://logs.openstack.org/68/197568/6/check/gate-devstack-dsvm-cells/bd4ce30/ : SUCCESS in 39m 43s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/6/check/gate-nova-tox-functional/381b66d/ : SUCCESS in 12m 50s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/6/check/gate-grenade-dsvm-partial-ncpu/82fc0b0/ : SUCCESS in 53m 26s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-ironic-pxe_ssh/e461218/ : SUCCESS in 55m 23s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-nova-v21-full/6702347/ : SUCCESS in 1h 04m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-cells/4578851/ : FAILURE in 1h 02m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-full-ceph/42e1063/ : SUCCESS in 1h 03m 19s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-multinode-full/76d3126/ : SUCCESS in 50m 32s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-multinode-smoke/173f7cc/ : SUCCESS in 47m 06s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/68/197568/6/check/gate-tempest-dsvm-neutron-multinode-smoke/79dd855/ : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c0a77be75f3e6e8bb43754a0e20ddff7e43b9d74","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-23 23:27:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/6/Nova-ML2-Sriov : SUCCESS in 1h 11m 25s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"65f10504ecd089e43a474e9cbdff8b8969f80f29","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-28 16:57:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/6 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"9bd1e509e55cb7e593987455b83dea1ab76e0927","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-08-11 12:36:02.000000000","message":"Patch Set 6: Code-Review-2\n\nSorry, you need to get a nova blueprint approved before we can merge this feature code.\nUnfortunately, as we have also now past the non-priority feature freeze for Liberty, you will need to submit this blueprint for Mitaka.\nFor more details on our process, please see:\nhttps://wiki.openstack.org/wiki/Nova/Liberty_Release_Schedule\nAny questions, please email me or find me on IRC johnthetubaguy","accounts_in_message":[],"_revision_number":6},{"id":"67eac3a7deaf2a745f3c4d5e81ce4729b360a9d6","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-09-22 10:00:03.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9ec54fd1ebfa08c289de619ef4e460eece44ac17","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-09-22 10:00:11.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"0b28fd7c3b7cc8702543aa1e5bea1b2b11061918","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-09-23 09:40:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"56b8867a8c007ce3a4243c21b74cee7b857f8ee0","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-09-23 13:30:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"935f62b202117b049715952ff71e4285bf6f7dd1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-23 14:30:24.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/7/check-ibm-tempest-dsvm-full/873a9e7 : SUCCESS in 55m 16s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/7/check-ibm-tempest-dsvm-postgres-full/0db015f : SUCCESS in 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e1ac3b2eec0abe151d31e81522bcc476b0585318","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-23 14:34:57.000000000","message":"Patch Set 7: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/7/12120/index.html : FAILURE in 57 m 48 s","accounts_in_message":[],"_revision_number":7},{"id":"365f6a2dd9ff00688f4299d96a41a60021cbd9ef","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-23 14:46:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/7/check/dsvm-tempest-xen/38eb457 : FAILURE in 1h 08m 30s","accounts_in_message":[],"_revision_number":7},{"id":"658f7eacbadd8a6ee0616551cbe18e7555ceaa29","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-23 14:55:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/7/check/gate-nova-pep8/79da61c/ : SUCCESS in 6m 44s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/7/check/gate-nova-docs/6d970eb//doc/build/html/ : SUCCESS in 11m 32s\n- gate-nova-python27 http://logs.openstack.org/68/197568/7/check/gate-nova-python27/9058788/ : FAILURE in 15m 20s\n- gate-nova-python34 http://logs.openstack.org/68/197568/7/check/gate-nova-python34/9a790ff/ : SUCCESS in 10m 38s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-full/2cca407/ : SUCCESS in 1h 06m 07s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-postgres-full/49119e8/ : SUCCESS in 59m 59s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-neutron-full/dc83480/ : SUCCESS in 1h 16m 11s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/7/check/gate-grenade-dsvm/8316c3c/ : SUCCESS in 35m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-large-ops/06536b1/ : SUCCESS in 32m 21s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-neutron-large-ops/817b89e/ : SUCCESS in 33m 01s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/7/check/gate-nova-tox-functional/aa9798b/ : SUCCESS in 18m 03s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/7/check/gate-grenade-dsvm-partial-ncpu/f45484e/ : SUCCESS in 41m 30s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/7e90c13/ : SUCCESS in 37m 51s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-nova-v20-api/dae89ab/ : SUCCESS in 31m 15s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-nova-v20-api-legacy/07a4644/ : SUCCESS in 32m 10s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-cells/5962c89/ : SUCCESS in 40m 44s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-full-ceph/7091cd1/ : SUCCESS in 43m 52s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/7/check/gate-tempest-dsvm-multinode-full/01cb78e/ : SUCCESS in 50m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a8b07a13641befcdd7e3b6d71314c6b57c936bf3","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-09-23 15:25:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3acf752ab8ac4a1308b2c382bbee1980426dd912","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-23 15:46:22.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"1fd3e01adff25288a079f6d1558c05480992e52b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-23 16:20:09.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/112eb04c547946cfbef3d4bad4157891/index.html : SUCCESS in 13m 20s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b5ae2d204bd048d48a21040e1e9ecc54/index.html : SUCCESS in 12m 47s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/af6495b2807e442a814be308346e27b3/index.html : SUCCESS in 15m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/7/check/real-db-upgrade_nova_percona_user_001:th-percona/a8945a49086247bb8502431c8e7e64bd/index.html : SUCCESS in 15m 12s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/797669a65ebd42c88f5091f612280efd/index.html : SUCCESS in 13m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/7/check/real-db-upgrade_nova_percona_user_002:th-percona/35bf969785d1496ebc1e119259a6d402/index.html : SUCCESS in 15m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"8fde340f5e1fe4525b49426df9ae852caf1924e6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-23 16:22:06.000000000","message":"Patch Set 8: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/8/12129/index.html : SUCCESS in 53 m 08 s","accounts_in_message":[],"_revision_number":8},{"id":"193dd40594d3fbbd200625a446e00d1b1758fd55","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-23 16:37:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/8/check/gate-nova-pep8/345e048/ : SUCCESS in 7m 41s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/8/check/gate-nova-docs/e4853a8//doc/build/html/ : SUCCESS in 14m 56s\n- gate-nova-python27 http://logs.openstack.org/68/197568/8/check/gate-nova-python27/b3f83a2/ : SUCCESS in 15m 57s\n- gate-nova-python34 http://logs.openstack.org/68/197568/8/check/gate-nova-python34/5e024cb/ : SUCCESS in 10m 50s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-full/53e8503/ : SUCCESS in 54m 15s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-postgres-full/43d60e4/ : SUCCESS in 56m 15s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-neutron-full/2d3a335/ : SUCCESS in 58m 56s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/8/check/gate-grenade-dsvm/fc7c222/ : SUCCESS in 39m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-large-ops/c042f94/ : SUCCESS in 34m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-neutron-large-ops/076f2e6/ : SUCCESS in 25m 03s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/8/check/gate-nova-tox-functional/6a319fc/ : SUCCESS in 13m 53s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/8/check/gate-grenade-dsvm-partial-ncpu/f48b6e3/ : SUCCESS in 39m 11s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0da889e/ : FAILURE in 33m 53s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-nova-v20-api/8e088dc/ : SUCCESS in 55m 20s (non-voting)\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-nova-v20-api-legacy/92bc1ab/ : SUCCESS in 38m 17s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-cells/bfa5e00/ : SUCCESS in 35m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-full-ceph/396cb24/ : SUCCESS in 47m 37s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/8/check/gate-tempest-dsvm-multinode-full/0321404/ : FAILURE in 1h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"24ef8b7ce744d425a6aebe30653bba9e8e80fa2e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-23 16:46:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/8/check/dsvm-tempest-xen/1b3ff55 : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":8},{"id":"1dc8092e92514bc25de284cb5a336ae89a0f8f56","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-23 17:14:38.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/197568/8/ : SUCCESS in 1h 44m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3414eb689b4a37501bc7363de777f04bcd766255","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-23 17:15:17.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/197568/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"4219bcc592dac8458489d625377f0c6e5a6fe6ae","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-09-23 17:15:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/68/197568/8/check/check-dsvm-vz7-tempest-exe-minimal/5c4be5e/ : SUCCESS in 28m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"389f84f5a7daa0d80fb2c9f324d376812636147e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-23 17:30:38.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/756185c5d6cb465e92bf4ab6031f50e1/index.html : SUCCESS in 12m 02s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/587aa4db986e4a2ea8734ddc5a4978b1/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d229b7b319b7452397be732bcd1701fa/index.html : SUCCESS in 12m 10s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/8/check/real-db-upgrade_nova_percona_user_001:th-percona/1cfde2e15f1845f1b2dfbce61dc32237/index.html : SUCCESS in 16m 43s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/2442e2373a0e41eaa10bc47945d71513/index.html : SUCCESS in 16m 27s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/8/check/real-db-upgrade_nova_percona_user_002:th-percona/b698e3cf45e34620840c1890fa83a043/index.html : SUCCESS in 14m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"2cb2c2e1cb6fe9d98f488621a258f31c848b5975","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-09-23 17:31:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/68/197568/8/check/dsvm-tempest-cinder/7a89f53/ : SUCCESS in 43m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"760dcdcf95824ff50aa392310796f269edeeaed6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-23 18:46:06.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/8/check-ibm-tempest-dsvm-full/f6c8b16 : SUCCESS in 55m 56s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/197568/8/check-ibm-tempest-dsvm-postgres-full/bd74594 : SUCCESS in 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f56a8e198ee092d3247ac48c3ff36ff825e7306c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-23 19:06:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/8/Nova-ML2-Sriov : FAILURE in 16m 52s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"277027c642559156aee88205b669455b2ab14dfc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-24 10:30:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/8 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"ed93960a170b8f33a569c299b1f50d9a79e2fe8b","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-09-29 09:35:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c48c815279b9e67cb622c8511be125685f0b8dba","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-09-29 10:33:41.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/68/197568/9/check/check-ibm-tempest-dsvm-full/5817bbb/ : SUCCESS in 55m 20s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/68/197568/9/check/check-ibm-tempest-dsvm-postgres-full/58cf1d6/ : SUCCESS in 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7e16f81729fa81087b84811a1c64c6d052db66e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-29 10:36:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/9/check/gate-nova-pep8/a00e88c/ : SUCCESS in 9m 24s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/9/check/gate-nova-docs/a113ac9//doc/build/html/ : SUCCESS in 14m 36s\n- gate-nova-python27 http://logs.openstack.org/68/197568/9/check/gate-nova-python27/cdcb7dd/ : SUCCESS in 15m 40s\n- gate-nova-python34 http://logs.openstack.org/68/197568/9/check/gate-nova-python34/b594b58/ : SUCCESS in 9m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-full/1de1ba3/ : SUCCESS in 55m 03s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-postgres-full/2313061/ : SUCCESS in 37m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-neutron-full/23463c6/ : FAILURE in 52m 54s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/9/check/gate-grenade-dsvm/8533935/ : SUCCESS in 41m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-large-ops/abd9c79/ : SUCCESS in 30m 18s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-neutron-large-ops/9535fa2/ : SUCCESS in 33m 15s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/9/check/gate-nova-tox-functional/e3fc34b/ : SUCCESS in 12m 32s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/9/check/gate-grenade-dsvm-partial-ncpu/ea1bb07/ : SUCCESS in 46m 02s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e3b5361/ : SUCCESS in 46m 28s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-nova-v20-api/0cbde79/ : SUCCESS in 33m 42s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-nova-v20-api-legacy/dea8ece/ : SUCCESS in 40m 42s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-cells/386914a/ : SUCCESS in 34m 02s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-full-ceph/be8cfe1/ : SUCCESS in 50m 51s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/9/check/gate-tempest-dsvm-multinode-full/8fba8f4/ : FAILURE in 46m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"76f67a056b6b1323e46e1c1aaedafdadd8a94305","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-09-29 10:45:12.000000000","message":"Patch Set 9:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1403291\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"0b3a7669f498ba0c6a062d4b79a28f271e1d9d7b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-09-29 10:48:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/9/check/dsvm-tempest-xen/642f789 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":9},{"id":"2df8be8b06a24fb55f7e4863feb046bc46570a36","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-09-29 10:50:59.000000000","message":"Patch Set 9: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/9/12539/index.html : SUCCESS in 01 h 11 m 34 s","accounts_in_message":[],"_revision_number":9},{"id":"85f2a4d54b5ce310b831888b6ad2cd3c006f28b4","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-09-29 11:08:19.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d119ca349bee4804a011574d20107214/index.html : SUCCESS in 14m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b6031bbb5e0547f5a7885f25e0bc8165/index.html : SUCCESS in 15m 47s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1f14d8a187cc49d582f1eea574e8f0f2/index.html : SUCCESS in 16m 06s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/9/check/real-db-upgrade_nova_percona_user_001:th-percona/0ffaa2146689460ea89304432bf7bdef/index.html : SUCCESS in 15m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d34dfed7f7274aa5a0d17b795357a041/index.html : SUCCESS in 12m 08s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/9/check/real-db-upgrade_nova_percona_user_002:th-percona/c1c143de5c28427c9f0c07d19219f97a/index.html : SUCCESS in 12m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"9d315f62a4dd759b5f47fa730b5f87c8a8c693cb","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-09-29 11:22:28.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/197568/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"84b4fd437fa44b5d758b50733b7ba7618c3fcb2f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-09-29 15:04:23.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/197568/9/ : FAILURE in 2h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4c6208dec30b4384588b9e06ec4ade622cda8275","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-09-29 15:53:00.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e52bf7437aa4a4626b8ee748cc00dc366da40477","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-09-29 17:51:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/9/Nova-ML2-Sriov : FAILURE in 13m 43s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"f58c888dd4e4839f64efcc6abaef0d9d2a801418","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-09-30 16:19:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/9 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"88f4ddb15ad8a50aa03c8e402dc0dcb08711f343","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-10-05 11:09:19.000000000","message":"Patch Set 9: -Code-Review\n\nblueprint now approved","accounts_in_message":[],"_revision_number":9},{"id":"4bf756fb9ccdae2de7ca5a3b8fde66a94764af18","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-10-05 11:27:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"06ef33297c35cfe3739639fdfa3e709edbdee21e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-05 12:22:59.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/68/197568/10/check/check-ibm-tempest-dsvm-full/6f1c1de/ : SUCCESS in 53m 46s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/68/197568/10/check/check-ibm-tempest-dsvm-postgres-full/cc42aa7/ : SUCCESS in 50m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"808dea51c4158eb78af51297793430128c2b95b7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-05 12:28:16.000000000","message":"Patch Set 10: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/10/12909/index.html : SUCCESS in 57 m 32 s","accounts_in_message":[],"_revision_number":10},{"id":"1e058333a8cb154822afa5a61abf0158c9977c80","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-05 12:31:04.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/197568/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"7faeb89aeef0c589e0e73361d630994437f5bea3","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-05 12:44:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/10/check/dsvm-tempest-xen/9ae142a : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":10},{"id":"3b7fe6c6dae5a4b582688f12eda642ea571d2dee","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-05 12:50:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/10/check/gate-nova-pep8/e80d9cb/ : SUCCESS in 6m 19s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/10/check/gate-nova-docs/d702555//doc/build/html/ : SUCCESS in 16m 06s\n- gate-nova-python27 http://logs.openstack.org/68/197568/10/check/gate-nova-python27/9b7235b/ : SUCCESS in 13m 39s\n- gate-nova-python34 http://logs.openstack.org/68/197568/10/check/gate-nova-python34/7bc7ab2/ : SUCCESS in 10m 08s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-full/d697dc2/ : SUCCESS in 57m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-postgres-full/473e945/ : SUCCESS in 42m 24s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-neutron-full/b8ab59e/ : SUCCESS in 1h 16m 55s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/10/check/gate-grenade-dsvm/0422067/ : SUCCESS in 45m 30s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-large-ops/acc3eb6/ : SUCCESS in 28m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-neutron-large-ops/b42fcfc/ : SUCCESS in 26m 03s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/10/check/gate-nova-tox-functional/1e5433a/ : SUCCESS in 17m 08s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/10/check/gate-grenade-dsvm-partial-ncpu/a1f648a/ : FAILURE in 46m 12s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/efb510a/ : SUCCESS in 46m 21s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-nova-v20-api/a4c9d27/ : SUCCESS in 32m 52s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-nova-v20-api-legacy/b51bccd/ : SUCCESS in 30m 41s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-cells/9c104bf/ : SUCCESS in 51m 24s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-full-ceph/d291aa4/ : SUCCESS in 47m 36s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-multinode-full/ba981a2/ : SUCCESS in 51m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3466b84f75ac5a553945a855d22ad929d42e8c9d","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-05 12:50:32.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-partial-ncpu: https://bugs.launchpad.net/bugs/1446583\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"8b5e30419db2236ccfbe8de642f683b8537f11a2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-10-05 14:35:47.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"70f5c969c05bb7c93394b819541bd0eb2dfc4e7c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-05 14:37:28.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/10/Nova-ML2-Sriov : SUCCESS in 54m 43s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"63beeecfda1e083e19687879dbf35547ae0ea831","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-05 14:40:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/10 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"4d4bde1422ea5f9d911d37a122936975c990e423","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-05 16:22:17.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/10/check/gate-nova-pep8/0116560/ : SUCCESS in 9m 33s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/10/check/gate-nova-docs/2797ddb//doc/build/html/ : SUCCESS in 9m 32s\n- gate-nova-python27 http://logs.openstack.org/68/197568/10/check/gate-nova-python27/64ddaa3/ : SUCCESS in 12m 40s\n- gate-nova-python34 http://logs.openstack.org/68/197568/10/check/gate-nova-python34/60678eb/ : SUCCESS in 8m 38s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-full/483f698/ : SUCCESS in 58m 51s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-postgres-full/17e73e0/ : SUCCESS in 1h 01m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-neutron-full/98b8f34/ : SUCCESS in 1h 21m 35s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/10/check/gate-grenade-dsvm/e8218e4/ : SUCCESS in 59m 58s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-large-ops/4cd18fe/ : SUCCESS in 45m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-neutron-large-ops/361085f/ : SUCCESS in 43m 59s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/10/check/gate-nova-tox-functional/aa7779a/ : SUCCESS in 17m 52s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/10/check/gate-grenade-dsvm-partial-ncpu/2ecd8d3/ : FAILURE in 55m 46s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/408ac80/ : SUCCESS in 56m 33s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-nova-v20-api/a42b962/ : SUCCESS in 52m 50s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-nova-v20-api-legacy/9578f3b/ : SUCCESS in 48m 50s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-cells/e8d6f24/ : SUCCESS in 56m 36s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-full-ceph/b0d32ac/ : SUCCESS in 1h 05m 36s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/10/check/gate-tempest-dsvm-multinode-full/d54601a/ : SUCCESS in 57m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3d577bd3478ad7baee7d0ca74dfa1266272bd88e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-05 17:01:26.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-partial-ncpu: https://bugs.launchpad.net/bugs/1446583\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"058f5cd294fb92200c2e5dbc432e76a4085f93c1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-05 18:52:32.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/197568/10/ : FAILURE in 2h 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bd28b859a616f96642c42d1637477059a58b88c9","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-10-07 12:43:08.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"839c7a7dc82b15e752478e859a1a470168b8cece","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-07 15:53:39.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/cc44c0190b0f4083a37de72a8311ca74/index.html : SUCCESS in 14m 11s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6b312f7f482c453f894d8039edbc811b/index.html : SUCCESS in 15m 37s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/bafb9de509124e0bbbc279f71577fa1a/index.html : SUCCESS in 15m 32s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/10/check/real-db-upgrade_nova_percona_user_001:th-percona/a2665da8507c40a28167e761c9071ecb/index.html : SUCCESS in 17m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/fc12f278cf4045e6bd168ddb221f91cd/index.html : SUCCESS in 13m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/10/check/real-db-upgrade_nova_percona_user_002:th-percona/535cac856acc42f182b5674cce7fa7b4/index.html : SUCCESS in 15m 59s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"3c1016c6a987cc1aa4f8bf6264949e8b30383d30","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-10-07 15:59:01.000000000","message":"Patch Set 10:\n\n(3 comments)\n\nThanks for your review I\u0027m not sure to have understand last comment, can you give me more information ?","accounts_in_message":[],"_revision_number":10},{"id":"8aa3d9ad5393c39cce34ef1054cda34cbdf549e8","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-10-08 09:30:11.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"09b66d0d951a5a6996ad7d8d2d06bc98de86195d","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-10-08 13:23:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d7dbf417fbe93ee4abce75b47331bc6dd8aca174","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-08 13:52:08.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/197568/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"16d4a8cf2c6a113db6845444b3ce5d28164b44c5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-08 14:23:08.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/92294c1f71fb4944b6bd42e2e388b1ef/index.html : SUCCESS in 14m 59s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7c9e3e5f7e644452b5f1ba6b426cbb60/index.html : SUCCESS in 21m 28s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4a3536bf299646c6865feb7045eac1c2/index.html : SUCCESS in 13m 30s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/11/check/real-db-upgrade_nova_percona_user_001:th-percona/1a810375fbc343438e3a3d323a920680/index.html : SUCCESS in 14m 20s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/8a71ad23451345d9af2d3f70e5098991/index.html : SUCCESS in 14m 11s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/11/check/real-db-upgrade_nova_percona_user_002:th-percona/05384a147867440587763e102c812b74/index.html : SUCCESS in 14m 00s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"bbd4f104eaa69435bd92456aa0bc0c83ce744911","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-08 14:27:15.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/68/197568/11/check/check-ibm-tempest-dsvm-full/d813d0b/ : SUCCESS in 1h 01m 28s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/68/197568/11/check/check-ibm-tempest-dsvm-postgres-full/2a712c1/ : SUCCESS in 53m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9af4b91b401db31f8408a4d855f9e1ddc910804c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-08 14:37:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/11/check/gate-nova-pep8/145d69a/ : SUCCESS in 7m 31s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/11/check/gate-nova-docs/4d195cc//doc/build/html/ : SUCCESS in 11m 45s\n- gate-nova-python27 http://logs.openstack.org/68/197568/11/check/gate-nova-python27/a5542fd/ : SUCCESS in 14m 16s\n- gate-nova-python34 http://logs.openstack.org/68/197568/11/check/gate-nova-python34/c6add11/ : SUCCESS in 12m 21s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-full/e7a31e3/ : SUCCESS in 40m 32s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-postgres-full/973534c/ : SUCCESS in 38m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-neutron-full/51e3523/ : SUCCESS in 1h 00m 54s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/11/check/gate-grenade-dsvm/03af50a/ : SUCCESS in 38m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-large-ops/502a6db/ : SUCCESS in 34m 36s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-neutron-large-ops/5eedb7b/ : SUCCESS in 37m 13s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/11/check/gate-nova-tox-functional/2ef15be/ : SUCCESS in 17m 41s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/11/check/gate-grenade-dsvm-partial-ncpu/70b858d/ : SUCCESS in 41m 23s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/39e61c5/ : SUCCESS in 43m 18s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-nova-v20-api/2dd1b12/ : SUCCESS in 52m 06s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-nova-v20-api-legacy/f0bd36c/ : SUCCESS in 42m 12s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-cells/a23fea2/ : FAILURE in 43m 29s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-full-ceph/c858738/ : SUCCESS in 53m 48s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/11/check/gate-tempest-dsvm-multinode-full/a289f26/ : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"563338253081604185b7b02d555550c7032b02b1","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-10-08 14:39:16.000000000","message":"Patch Set 11:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells: https://bugs.launchpad.net/bugs/1466696\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":11},{"id":"d81add2fd1aa904465978f03abdd351548bcca92","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-08 14:40:08.000000000","message":"Patch Set 11: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/11/13223/index.html : SUCCESS in 01 h 11 m 18 s","accounts_in_message":[],"_revision_number":11},{"id":"154841659e3d96030a6bbcbd9de0292b020cd7a4","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-08 14:45:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/11/check/dsvm-tempest-xen/d13424a : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":11},{"id":"11d101fb172fc65a23d6dc813ddfbfaf4df3f6ff","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-08 15:01:48.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/197568/11/ : SUCCESS in 1h 36m 54s","accounts_in_message":[],"_revision_number":11},{"id":"026d3a33a533b5592c00e0804f657dbd00528f7d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-08 18:35:49.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/11/Nova-ML2-Sriov : FAILURE in 18m 25s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"596f4ac8d2ef80a5db8562d410cbc6e5d3ff5d7c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-09 01:49:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/11 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"cd6953f6e67d79e3a40b73a01c88de091a591e33","author":{"_account_id":8802,"name":"Vladik Romanovsky","email":"vromanso@redhat.com","username":"vladikr"},"date":"2015-10-09 15:06:45.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d236df7ffe1a78109a3f26f3ad139fa8813fb2c9","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-11-03 05:44:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ad6c7a891e49a18774e8fec38c35b3366ef68e50","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-03 06:11:32.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6de1de6d9b304c4a8d5ea73697504e07/index.html : SUCCESS in 16m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4e30b0b4f115447285cbb7a7a60ad4f4/index.html : SUCCESS in 13m 07s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/274ef350944643419e3796d265b72f7c/index.html : SUCCESS in 13m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/12/check/real-db-upgrade_nova_percona_user_001:th-percona/dace06f5e3d54d79b6fdc08a458d8a7d/index.html : SUCCESS in 16m 28s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/68/197568/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/238b07fbe1624b2a91aae858fce71a54/index.html : SUCCESS in 14m 01s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/68/197568/12/check/real-db-upgrade_nova_percona_user_002:th-percona/a8e3d3a7d383434dbd8d8a5fd0844e37/index.html : SUCCESS in 15m 30s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"02249934864d06edde1b7178e2f9817ddba9910b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-03 06:16:36.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/197568/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"05fceccb48bf62b64fb2b6aac43541d531de6dce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-11-03 06:41:32.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/68/197568/12/check/check-ibm-tempest-dsvm-full/5eaf84e/ : SUCCESS in 54m 46s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/68/197568/12/check/check-ibm-tempest-dsvm-postgres-full/eb84d3b/ : SUCCESS in 52m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bee01c46293f19d6add52a204719ac1991f378ad","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-03 06:43:09.000000000","message":"Patch Set 12: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/68/197568/12/14759/index.html : SUCCESS in 55 m 23 s","accounts_in_message":[],"_revision_number":12},{"id":"cb76d79e5eb295ff8e3a25ae2d436ef77bb4ff2a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-03 06:46:40.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/197568/12/ : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":12},{"id":"c0e539f919fbf357e53ce5f18d5fe8704c548ddb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-03 06:58:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/12/check/gate-nova-pep8/d100809/ : SUCCESS in 8m 18s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/12/check/gate-nova-docs/9ae2283//doc/build/html/ : SUCCESS in 14m 09s\n- gate-nova-python27 http://logs.openstack.org/68/197568/12/check/gate-nova-python27/d094967/ : SUCCESS in 10m 56s\n- gate-nova-python34 http://logs.openstack.org/68/197568/12/check/gate-nova-python34/470199a/ : SUCCESS in 10m 19s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-full/fb5c027/ : SUCCESS in 1h 00m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-postgres-full/7f5c12d/ : FAILURE in 55m 16s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-neutron-full/e6795e9/ : SUCCESS in 1h 11m 24s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/12/check/gate-grenade-dsvm/08c5272/ : SUCCESS in 57m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-large-ops/026280b/ : SUCCESS in 35m 44s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-neutron-large-ops/c147f7b/ : SUCCESS in 41m 28s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/12/check/gate-nova-tox-functional/8935c4e/ : SUCCESS in 13m 25s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/12/check/gate-grenade-dsvm-partial-ncpu/3f50fa1/ : SUCCESS in 1h 02m 47s\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/68/197568/12/check/gate-grenade-dsvm-multinode-nv/b3ef954/ : SUCCESS in 58m 22s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8b680db/ : SUCCESS in 47m 45s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-nova-v20-api/ae18a4e/ : SUCCESS in 40m 39s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-nova-v20-api-legacy/90cd32b/ : SUCCESS in 42m 56s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-cells/3378eec/ : SUCCESS in 53m 59s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-full-ceph/c329df9/ : SUCCESS in 58m 39s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-multinode-full/ed2c387/ : SUCCESS in 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5da8425a68a7dcc67966069005ab3bcf47c48610","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-11-03 06:58:35.000000000","message":"Patch Set 12:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1501366\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":12},{"id":"fa99c91d30b20dd81fe940f9cfdebb6721925963","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-03 07:07:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/68/197568/12/check/dsvm-tempest-xen/ac4dbf5 : SUCCESS in 1h 16m 44s","accounts_in_message":[],"_revision_number":12},{"id":"ef28ac368ee818053c5ffa091de1723f756f003f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-11-03 07:33:30.000000000","message":"Patch Set 12:\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/68/197568/12/compute-nfv-flavors : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"1476a29927006f877424d490cd3bcf585d8f00c9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-03 08:46:38.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/12/Nova-ML2-Sriov : FAILURE in 26m 03s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"380adaed18c361070ff2d504f83fcfb81bef4d2e","author":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"date":"2015-11-03 10:41:52.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c6b2720e892fa693d05aba5f3a500ec93ea9ec55","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-04 14:17:33.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/197568/12/Nova-ML2-Sriov : SUCCESS in 1h 09m 41s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"67eb0537c399e83b98546c2d43bee8cf50cc42d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-05 18:24:57.000000000","message":"Patch Set 12: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/197568/12 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/197568/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"0aedefc8c08ea2a5340327b02632420f5eb52daa","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-11-06 09:02:23.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"f31343db8b3adc73d4cf5d250fda3191694d9779","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-06 10:02:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/68/197568/12/check/gate-nova-pep8/314316f/ : SUCCESS in 7m 19s\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/12/check/gate-nova-docs/2102113//doc/build/html/ : SUCCESS in 14m 21s\n- gate-nova-python27 http://logs.openstack.org/68/197568/12/check/gate-nova-python27/ec42ff7/ : SUCCESS in 15m 24s\n- gate-nova-python34 http://logs.openstack.org/68/197568/12/check/gate-nova-python34/70bc477/ : SUCCESS in 9m 56s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-full/d8ca5d5/ : SUCCESS in 42m 34s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-postgres-full/6f30cdd/ : SUCCESS in 40m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-neutron-full/5c157f1/ : SUCCESS in 58m 06s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/12/check/gate-grenade-dsvm/240a33c/ : SUCCESS in 44m 07s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-large-ops/456faeb/ : SUCCESS in 31m 12s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-neutron-large-ops/2fc6a84/ : SUCCESS in 26m 43s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/12/check/gate-nova-tox-functional/6e4f8de/ : SUCCESS in 12m 36s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/68/197568/12/check/gate-grenade-dsvm-partial-ncpu/7fe3c12/ : SUCCESS in 42m 37s\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/68/197568/12/check/gate-grenade-dsvm-multinode-nv/7708847/ : FAILURE in 32m 31s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a2b6635/ : SUCCESS in 42m 25s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-nova-v20-api/b1fcf31/ : SUCCESS in 41m 38s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-nova-v20-api-legacy/c5f5843/ : SUCCESS in 42m 42s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-cells/4fa3d14/ : SUCCESS in 58m 23s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-full-ceph/6e55132/ : SUCCESS in 47m 44s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/68/197568/12/check/gate-tempest-dsvm-multinode-full/89183a0/ : FAILURE in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9ccf7b6cacf39f275161a920fdde9cd61b24a68e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-06 10:39:22.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/197568/12/ : FAILURE in 1h 36m 23s","accounts_in_message":[],"_revision_number":12},{"id":"2f99b17660273fbea7776ef48b9cba0d0ab90cd5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-11-06 10:53:09.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/68/197568/12/check/dsvm-tempest-cinder/8e3dfa6/ : SUCCESS in 49m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"bcaaab829487a0d44558e66eac35fdd719dfeaae","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-11-06 10:57:25.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/68/197568/12/check/check-dsvm-vz7-tempest-exe-minimal/cc75fc5/ : FAILURE in 51m 54s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"e8530e7ed63033aba7c7efa71f3ccb662894b8c8","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-11-19 12:00:41.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ce9bae7c66f064e45d367672f90cd161c5477e16","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-19 12:00:49.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"e37924861c43b95df6cda41b4e20850d573ee0cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-19 14:31:31.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/68/197568/12/gate/gate-nova-docs/dadbab3//doc/build/html/ : SUCCESS in 9m 45s\n- gate-nova-pep8 http://logs.openstack.org/68/197568/12/gate/gate-nova-pep8/140a1bd/ : SUCCESS in 9m 31s\n- gate-nova-python27 http://logs.openstack.org/68/197568/12/gate/gate-nova-python27/68977cd/ : SUCCESS in 14m 07s\n- gate-nova-python34 http://logs.openstack.org/68/197568/12/gate/gate-nova-python34/7564c99/ : SUCCESS in 12m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-full/4de78e0/ : SUCCESS in 58m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-postgres-full/a52cc1e/ : SUCCESS in 54m 47s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-neutron-full/53d2fe8/ : SUCCESS in 1h 11m 52s\n- gate-grenade-dsvm http://logs.openstack.org/68/197568/12/gate/gate-grenade-dsvm/8e4f70d/ : SUCCESS in 46m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-large-ops/3d7d106/ : SUCCESS in 36m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-neutron-large-ops/494e3eb/ : SUCCESS in 34m 51s\n- gate-nova-tox-functional http://logs.openstack.org/68/197568/12/gate/gate-nova-tox-functional/46c93bf/ : SUCCESS in 17m 03s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/68/197568/12/gate/gate-grenade-dsvm-multinode/0696e23/ : SUCCESS in 48m 16s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-full-ceph/cc4bc96/ : SUCCESS in 57m 50s\n- gate-tempest-dsvm-cells http://logs.openstack.org/68/197568/12/gate/gate-tempest-dsvm-cells/d4d89be/ : SUCCESS in 47m 51s","accounts_in_message":[],"_revision_number":12},{"id":"66b21985f591d109364eeeba18a1bfcd523f5b5c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-19 14:31:35.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"e388ff8ff52b909c10eb894223843a3a5e0787ea","revisions":{"24d362c2d35270aa4d10377b776e4e66c6b2f529":{"kind":"REWORK","_number":1,"created":"2015-07-01 13:03:51.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/1"}}},"commit":{"parents":[{"commit":"cdf3cda243e2009aef50bde0baf0c9ae7ec01150","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdf3cda243e2009aef50bde0baf0c9ae7ec01150"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-06-30 14:14:25.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-07-01 13:03:14.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-realtime\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24d362c2d35270aa4d10377b776e4e66c6b2f529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24d362c2d35270aa4d10377b776e4e66c6b2f529"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cdf3cda243e2009aef50bde0baf0c9ae7ec01150","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"33fd5bd5f76735b0dc6b7d157cfbebf04c17e874":{"kind":"REWORK","_number":2,"created":"2015-07-01 13:53:54.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/2"}}},"commit":{"parents":[{"commit":"096b37198e5fa5df332ffbc2cef7562d63b772be","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/096b37198e5fa5df332ffbc2cef7562d63b772be"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-06-30 14:14:25.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-07-01 13:53:26.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33fd5bd5f76735b0dc6b7d157cfbebf04c17e874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33fd5bd5f76735b0dc6b7d157cfbebf04c17e874"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"096b37198e5fa5df332ffbc2cef7562d63b772be","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"603c8a6a1abd6cd280b3a475f1209b4b0b53c6a7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2015-07-09 10:03:27.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/3"}}},"commit":{"parents":[{"commit":"a6f5cafd09b5f494db6d7df1962ada1a590facca","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6f5cafd09b5f494db6d7df1962ada1a590facca"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-06-30 14:14:25.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-07-09 08:28:38.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/603c8a6a1abd6cd280b3a475f1209b4b0b53c6a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/603c8a6a1abd6cd280b3a475f1209b4b0b53c6a7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a6f5cafd09b5f494db6d7df1962ada1a590facca","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f44802e4ea37a54b118a3e58350e50a1dac2af2b":{"kind":"REWORK","_number":4,"created":"2015-07-15 12:50:50.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/4"}}},"commit":{"parents":[{"commit":"e71772b1bb073e1dddfdc51de90acb52efe7a817","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e71772b1bb073e1dddfdc51de90acb52efe7a817"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-06-30 14:14:25.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-07-15 12:49:07.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f44802e4ea37a54b118a3e58350e50a1dac2af2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f44802e4ea37a54b118a3e58350e50a1dac2af2b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e71772b1bb073e1dddfdc51de90acb52efe7a817","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c33c465b00a92492e33271b785d49c7b27c48330":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2015-07-21 15:22:42.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/5"}}},"commit":{"parents":[{"commit":"04ce536897c3c3e0f995cfa8cf02498db2c00c2b","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04ce536897c3c3e0f995cfa8cf02498db2c00c2b"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-06-30 14:14:25.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-07-21 09:59:25.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c33c465b00a92492e33271b785d49c7b27c48330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c33c465b00a92492e33271b785d49c7b27c48330"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"04ce536897c3c3e0f995cfa8cf02498db2c00c2b","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d8a4a644776c8b337147cd92a7274b3b30a66d6f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2015-07-23 10:47:19.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/6"}}},"commit":{"parents":[{"commit":"923272e6297fa3b1801c1e8f795ac8d7c0dd0058","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/923272e6297fa3b1801c1e8f795ac8d7c0dd0058"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-06-30 14:14:25.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-07-23 10:19:43.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8a4a644776c8b337147cd92a7274b3b30a66d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8a4a644776c8b337147cd92a7274b3b30a66d6f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"923272e6297fa3b1801c1e8f795ac8d7c0dd0058","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"256030c0dec9b773d66e1b3a65d8110c8d921247":{"kind":"REWORK","_number":7,"created":"2015-09-23 13:30:42.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/7"}}},"commit":{"parents":[{"commit":"7d1b391d871291eec531a0594ec7079c4ae96ee2","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d1b391d871291eec531a0594ec7079c4ae96ee2"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 09:47:37.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 13:25:06.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/256030c0dec9b773d66e1b3a65d8110c8d921247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/256030c0dec9b773d66e1b3a65d8110c8d921247"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7d1b391d871291eec531a0594ec7079c4ae96ee2","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"66c904c9892e9775b45b8e1178125bd25d28f8da":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2015-09-23 15:25:13.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/8"}}},"commit":{"parents":[{"commit":"7532ea6840bd2ef989092b74ee5e1de0721a8f29","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7532ea6840bd2ef989092b74ee5e1de0721a8f29"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 09:47:37.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 15:23:54.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66c904c9892e9775b45b8e1178125bd25d28f8da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66c904c9892e9775b45b8e1178125bd25d28f8da"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7532ea6840bd2ef989092b74ee5e1de0721a8f29","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"46bd1f1925ddff72fcb1fe01212775d99f9eec87":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2015-09-29 09:35:05.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/9"}}},"commit":{"parents":[{"commit":"bb9c9e20a8c88543677d2e3a8ea6628382fff6c7","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb9c9e20a8c88543677d2e3a8ea6628382fff6c7"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 09:47:37.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-29 09:31:11.000000000","tz":-240},"subject":"virt: add contraint to handle realtime policy","message":"virt: add contraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of the new introduced option cpu_realtime.\nIt verify if the configuration is good or raise the an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46bd1f1925ddff72fcb1fe01212775d99f9eec87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46bd1f1925ddff72fcb1fe01212775d99f9eec87"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bb9c9e20a8c88543677d2e3a8ea6628382fff6c7","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0dd300a4ea42c0ab29d5f40293c8cc9616b2beb6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":10,"created":"2015-10-05 11:27:11.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/10"}}},"commit":{"parents":[{"commit":"5552112cb213ada213bce1be0d95160c35dd451a","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5552112cb213ada213bce1be0d95160c35dd451a"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 09:47:37.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-10-05 10:19:36.000000000","tz":-240},"subject":"virt: add constraint to handle realtime policy","message":"virt: add constraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of new introduced option cpu_realtime.\nIt verifiy if the configuration is good or raise an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dd300a4ea42c0ab29d5f40293c8cc9616b2beb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dd300a4ea42c0ab29d5f40293c8cc9616b2beb6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5552112cb213ada213bce1be0d95160c35dd451a","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f6f2b1c3ba4fcf26286182a1cf009334d92241da":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2015-10-08 13:23:30.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/11"}}},"commit":{"parents":[{"commit":"b16aa88c45c1e2618d165a7a14fff6e1ed392b1c","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b16aa88c45c1e2618d165a7a14fff6e1ed392b1c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 09:47:37.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-10-08 13:19:27.000000000","tz":-240},"subject":"virt: add constraint to handle realtime policy","message":"virt: add constraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of new introduced option cpu_realtime.\nIt verifiy if the configuration is good or raise an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6f2b1c3ba4fcf26286182a1cf009334d92241da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6f2b1c3ba4fcf26286182a1cf009334d92241da"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b16aa88c45c1e2618d165a7a14fff6e1ed392b1c","is_merged_in_target_branch":false,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e388ff8ff52b909c10eb894223843a3a5e0787ea":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2015-11-03 05:44:31.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/68/197568/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/197568/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/197568/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/197568/12"}}},"commit":{"parents":[{"commit":"d61874d39502c09dc13052f3a7b16e4eece58a40","subject":"libvirt: add cpu schedular priority config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d61874d39502c09dc13052f3a7b16e4eece58a40"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-09-23 09:47:37.000000000","tz":-240},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@redhat.com","date":"2015-11-03 05:42:28.000000000","tz":-300},"subject":"virt: add constraint to handle realtime policy","message":"virt: add constraint to handle realtime policy\n\nThis commit makes pinning constraints to take into\naccount of new introduced option cpu_realtime.\nIt verifiy if the configuration is good or raise an\nerror to user.\n\nChange-Id: Iefbb60fcc4f8d1a4b6f8ca8bac9c9a0d8bec3a62\nBlueprint: libvirt-real-time\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e388ff8ff52b909c10eb894223843a3a5e0787ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e388ff8ff52b909c10eb894223843a3a5e0787ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d61874d39502c09dc13052f3a7b16e4eece58a40","is_merged_in_target_branch":true,"change_id":"I2b2cdebb28b2e452d64924e0f5da93b49b2ab6cb","change_number":197567,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
