)]}'
{"id":"openstack%2Fnova~202436","triplet_id":"openstack%2Fnova~master~I5dc647ef707ff04ad1cf7e788268cafb89f8ef03","project":"openstack/nova","branch":"master","topic":"bug/1417975","hashtags":[],"change_id":"I5dc647ef707ff04ad1cf7e788268cafb89f8ef03","subject":"neutron - race conditions in security-group additions","status":"ABANDONED","created":"2015-07-16 08:31:11.000000000","updated":"2015-08-26 23:48:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e90d5a2d6ff4f50650001b14acfacb1c13f912ab","_number":202436,"virtual_id_number":202436,"owner":{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2015-07-23 05:18:22.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2015-07-16 09:43:52.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2015-07-16 12:45:35.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-08-17 16:19:46.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2015-07-16 10:33:57.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2015-07-18 11:33:13.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2015-07-16 09:51:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2015-07-16 08:53:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"date":"2015-07-18 04:26:14.000000000","_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},{"value":-1,"date":"2015-08-03 14:27:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2015-07-17 08:29:42.000000000","_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"date":"2015-08-03 14:19:11.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-08-03 14:47:16.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-07-16 08:53:10.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-16 09:43:52.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-16 09:51:05.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-16 10:33:57.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-16 12:45:35.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-17 08:29:42.000000000","updated_by":{"_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},"reviewer":{"_account_id":12839,"name":"c. y. lee","username":"cylee","inactive":true},"state":"REVIEWER"},{"updated":"2015-07-18 11:33:13.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-23 05:18:22.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-03 14:19:11.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2015-08-03 14:27:39.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2015-08-17 16:19:46.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"76eaf0eb1a44dffa48a14dffb8d8b6e984d22847","author":{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},"date":"2015-07-16 08:31:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b6a1640d2f890c94cd326808d4678b92dc41120","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-07-16 08:53:10.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/36/202436/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5ca2622d7e3a48dd9fb3f270e2b14b40/index.html : SUCCESS in 11m 51s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/36/202436/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1c35dbb1ecf04af69352b92910979ac1/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/36/202436/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8b185abfb13a43b59866b850bfb61540/index.html : SUCCESS in 13m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/36/202436/1/check/real-db-upgrade_nova_percona_user_001:th-percona/3f7d55bded4c4ed8822854cdf98e6f37/index.html : SUCCESS in 15m 22s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/36/202436/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4ac5aab0452445cbb64874051c0210e9/index.html : SUCCESS in 13m 59s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/36/202436/1/check/real-db-upgrade_nova_percona_user_002:th-percona/d9ef363cc1824426a7a3ca7998fa29e4/index.html : SUCCESS in 14m 11s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"22020527e9eccb322c009b0379fedb956ff9e492","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-07-16 09:43:52.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/36/202436/1/4434/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"2957d8610eac9cb11f97ff27c02e5c694bdcd5e1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-16 09:45:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/36/202436/1/check/dsvm-tempest-xen/efe9b08 : FAILURE in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eb8094172dd36462e0f7b3b268455bc04ddbdbe9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-07-16 09:51:05.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/202436/1/check-ibm-tempest-dsvm-full/f9888c8 : SUCCESS in 1h 19m 30s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/202436/1/check-ibm-tempest-dsvm-postgres-full/493fbd6 : SUCCESS in 1h 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"683b2824bdba6ead2eb7fee6078380d7a20e9a28","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-16 10:07:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/36/202436/1/check/gate-nova-pep8/845198e/ : SUCCESS in 8m 25s\n- gate-nova-docs http://docs-draft.openstack.org/36/202436/1/check/gate-nova-docs/e8f00a7//doc/build/html/ : SUCCESS in 8m 46s\n- gate-nova-python27 http://logs.openstack.org/36/202436/1/check/gate-nova-python27/e2f6ec3/ : SUCCESS in 13m 37s\n- gate-nova-python34 http://logs.openstack.org/36/202436/1/check/gate-nova-python34/5d2d3b1/ : SUCCESS in 12m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-full/3974aad/ : SUCCESS in 39m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-postgres-full/afc8f50/ : SUCCESS in 50m 56s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-neutron/1074943/ : SUCCESS in 42m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-neutron-full/21a0c0b/ : SUCCESS in 59m 58s\n- gate-grenade-dsvm http://logs.openstack.org/36/202436/1/check/gate-grenade-dsvm/1cbaa23/ : SUCCESS in 49m 15s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-large-ops/8acd5f4/ : SUCCESS in 35m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-neutron-large-ops/1416890/ : SUCCESS in 31m 58s\n- gate-devstack-dsvm-cells http://logs.openstack.org/36/202436/1/check/gate-devstack-dsvm-cells/6fd703e/ : SUCCESS in 39m 15s\n- gate-nova-tox-functional http://logs.openstack.org/36/202436/1/check/gate-nova-tox-functional/5a60961/ : SUCCESS in 12m 22s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/36/202436/1/check/gate-grenade-dsvm-partial-ncpu/60bfef6/ : SUCCESS in 48m 27s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-ironic-pxe_ssh/f831cd2/ : SUCCESS in 43m 06s\n- gate-tempest-dsvm-docker http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-docker/1b1d1ce/ : FAILURE in 21m 20s (non-voting)\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-nova-v21-full/df3a657/ : SUCCESS in 1h 29m 12s\n- gate-tempest-dsvm-cells http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-cells/ca6d4d0/ : SUCCESS in 1h 22m 38s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-full-ceph/9165ffc/ : SUCCESS in 52m 04s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-multinode-full/e7fcc58/ : FAILURE in 48m 48s (non-voting)\n- gate-tempest-dsvm-multinode-smoke http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-multinode-smoke/9a393fa/ : SUCCESS in 1h 19m 38s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-smoke http://logs.openstack.org/36/202436/1/check/gate-tempest-dsvm-neutron-multinode-smoke/f84861b/ : SUCCESS in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac29848a8a7e7e11e82415ee03dc00703cd9e67c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-07-16 10:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/202436/1/Nova-ML2-Sriov : SUCCESS in 53m 47s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd1a0182140c02aacacf804867fea8a7b697afa2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-07-16 12:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- hyper-v-ci http://64.119.130.115/202436/1/ : SUCCESS in 1h 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"41c57f727e1cc866b568238332b9469bda765117","author":{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},"date":"2015-07-18 04:26:14.000000000","message":"Patch Set 1:\n\nxen: recheck","accounts_in_message":[],"_revision_number":1},{"id":"5694a5bc282a5001811c51189938dd6964b465f4","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-07-18 11:33:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/36/202436/1/check/dsvm-tempest-xen/8c8d63a : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"976e912d66f8bb7caca32df2abb79af3da4a799f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-07-23 05:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/202436/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/202436/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"5488a88d796d8770c655bb970bca5c342e899ca6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2015-08-03 14:27:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe766522cfaa85071481da677bc99b80426ba1cf","author":{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},"date":"2015-08-03 14:47:16.000000000","message":"Patch Set 1: Workflow-1\n\nAdd lock at api is not grate idea. Just like Dan Smith said. It is adaptive to add neutron api which can add/remove security group instead of setting full data about security group.","accounts_in_message":[],"_revision_number":1},{"id":"b7aba3779ecb988ffb0dc002bb9d71f8748a535e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-17 16:19:46.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"2d7e0a73e4e0fb929e6a292727579cc499575ad3","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2015-08-26 23:48:41.000000000","message":"Abandoned\n\nThis patch has been stalled for quite a while, so I am going to abandon it to keep the code review queue sane. Please restore the change when it is ready for review.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91e84315215a7a368b38ba8043fd8af787b368b3","revisions":{"91e84315215a7a368b38ba8043fd8af787b368b3":{"kind":"REWORK","_number":1,"created":"2015-07-16 08:31:11.000000000","uploader":{"_account_id":14447,"name":"lyanchih","email":"lyan.h@inwinstack.com","username":"lyanchih"},"ref":"refs/changes/36/202436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/36/202436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/202436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/202436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/202436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/36/202436/1"}}},"commit":{"parents":[{"commit":"3ba9137ac547e9e5aff23cb65b7b9e54d397ef1e","subject":"Merge \"Cleanup quota_class unittest with appropriate request context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ba9137ac547e9e5aff23cb65b7b9e54d397ef1e"}]}],"author":{"name":"lyanchih","email":"lyan.h@inwinstack.com","date":"2015-07-15 17:08:10.000000000","tz":0},"committer":{"name":"lyanchih","email":"lyan.h@inwinstack.com","date":"2015-07-16 08:30:28.000000000","tz":0},"subject":"neutron - race conditions in security-group additions","message":"neutron - race conditions in security-group additions\n\nWhen attempting to connect multiple security groups to a single server\nconcurrently, some of the security groups may not actually get connected\ndespite the call returning successfully and with no error.\n\nHow it happen:\nWhen nova-api want to add/remove security-group, it will send neutron\nlist-port request. It will response server\u0027s ports and their security\ngroup. When two request coming and they send list-port api at same\ntime. They will receive empty security-group and append\nsecurity-group-id. Finally send update_port with security-group array\nto neutron. Then last update will override previous update. Therefore\nyou will sometimes saw one of two secgroup or all of them.\n\nOther scenario:\nWhen you send add new secgroup and remove exist secgroup in same\ntime. It would probably cause those secgroup are all exist.\n\nAdd utils.synchronized to add_to_instance and remove_from_instance\n\nChange-Id: I5dc647ef707ff04ad1cf7e788268cafb89f8ef03\nCloses-Bug: 1417975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91e84315215a7a368b38ba8043fd8af787b368b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91e84315215a7a368b38ba8043fd8af787b368b3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3ba9137ac547e9e5aff23cb65b7b9e54d397ef1e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
