)]}'
{"id":"openstack%2Fnova~209812","triplet_id":"openstack%2Fnova~master~I53d9c327cb902dc150eac3b9b672d4a6f89c7fcc","project":"openstack/nova","branch":"master","topic":"pci_passthrough","hashtags":[],"change_id":"I53d9c327cb902dc150eac3b9b672d4a6f89c7fcc","subject":"use new oslo.versionedobjects","status":"ABANDONED","created":"2015-08-06 08:17:27.000000000","updated":"2015-08-08 12:20:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8026cd2bc6f887ffb5f400600fe99717fb3cd7c4","_number":209812,"virtual_id_number":209812,"owner":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2015-08-06 10:24:43.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2015-08-06 11:40:32.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-08-06 10:35:45.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2015-08-07 01:05:47.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2015-08-06 10:29:44.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2015-08-06 09:56:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2015-08-06 09:34:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-08-06 09:15:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-06 09:15:39.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2015-08-06 09:34:12.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-06 09:56:35.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-06 10:24:43.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-06 10:29:44.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-06 10:35:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-08-06 11:40:32.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-07 01:05:47.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d711ca8c0c8d953d873e5e9364aa02ee589c9d0e","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2015-08-06 08:17:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2534470d46e2f516ea0918b4cb94c80f4a334b28","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-06 08:34:02.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/12/209812/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/68e638c6cdda44d3b87fb7fda739174e/index.html : SUCCESS in 11m 48s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/12/209812/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5f778d11509a4f4997c0eeb2730755b6/index.html : SUCCESS in 15m 46s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/12/209812/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/40b6ad353b5d495d82856a6b096dd2ef/index.html : SUCCESS in 12m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/12/209812/1/check/real-db-upgrade_nova_percona_user_001:th-percona/b117c39a2da740168d5359f705e6e755/index.html : SUCCESS in 13m 25s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/12/209812/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ae639f0c98654ef1baa21aac5d986de9/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/12/209812/1/check/real-db-upgrade_nova_percona_user_002:th-percona/d94cc8a1e92c44c1ba1952cfc15076b6/index.html : SUCCESS in 13m 55s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4b3eb7754f159615482f1ae51d6e64f1fccdf7c","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-08-06 08:39:18.000000000","message":"Patch Set 1: Code-Review-1\n\nIf this is the only problem:\nhttp://paste.openstack.org/show/410220/\n\nI think you need to upgrade to include this fix:\nhttps://github.com/openstack/oslo.versionedobjects/commit/58070642b52a2332605ab5d834b7de323da9ba01\n\nWe might need to bump requirements, but that would be the better change here.","accounts_in_message":[],"_revision_number":1},{"id":"0a89e6837877c7822edd540f51eb354b650b5e48","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-08-06 08:46:28.000000000","message":"Patch Set 1:\n\nSo I think the best fix here is to bump the min version of oslo versioned objects:\nhttps://github.com/openstack/requirements/blob/master/global-requirements.txt#L91\n\nSeems like we need 0.6.0 or higher to make everything work","accounts_in_message":[],"_revision_number":1},{"id":"5a144a24f7991fe8d0be22bdb362faf9de3dbf7f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-06 09:06:17.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/209812/1/check-ibm-tempest-dsvm-full/9f9323c : SUCCESS in 48m 22s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/209812/1/check-ibm-tempest-dsvm-postgres-full/77cd8d2 : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf4ead694db03d8e44ebb97811762ff6e2f5eebf","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2015-08-06 09:08:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee283e52ca82cf9d5bbb043e30bb3f16164d9b79","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2015-08-06 09:15:39.000000000","message":"Patch Set 2: Code-Review-1\n\nI believe you need to update the global requirements, and you need to reference the bug there, I think.\n\nSee the link I gave in the previous comment:\nhttps://github.com/openstack/requirements/blob/master/global-requirements.txt#L91","accounts_in_message":[],"_revision_number":2},{"id":"2493bd79a937f00653f752d04214cd4e144e0652","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-06 09:34:12.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/12/209812/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/02ced84ab3f64fc2968ba53eb4a98532/index.html : SUCCESS in 11m 47s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/12/209812/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b19b2895d7b942e484ded36b618b0c0a/index.html : SUCCESS in 16m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/12/209812/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/09187d2b407c402285ec701b6e7ed35a/index.html : SUCCESS in 13m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/12/209812/2/check/real-db-upgrade_nova_percona_user_001:th-percona/bff56bbefa4a49e9a71b00cd929a11e5/index.html : SUCCESS in 15m 24s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/12/209812/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/200f4b9d743e4432bb5fa66575d64ffe/index.html : SUCCESS in 12m 12s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/12/209812/2/check/real-db-upgrade_nova_percona_user_002:th-percona/1b49d306902042c8827069214323df64/index.html : SUCCESS in 12m 20s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"6927941736c60b9619be8bd82edee998c646f104","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-08-06 09:56:35.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/209812/2/check-ibm-tempest-dsvm-full/3d732b3 : SUCCESS in 47m 25s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/209812/2/check-ibm-tempest-dsvm-postgres-full/fc396f7 : SUCCESS in 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57299658a3e1aa3b9a3ce10ad0dae0f582a1346b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-06 10:24:43.000000000","message":"Patch Set 2: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/12/209812/2/7279/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":2},{"id":"f17b3d2df936965943d9946df277f76138b74f0c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-06 10:29:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/12/209812/2/check/dsvm-tempest-xen/41731b0 : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c9fd1c319770d72f2c94a7789b0683ee8657012a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-06 10:35:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/12/209812/2/check/gate-nova-pep8/3d33177/ : SUCCESS in 6m 49s\n- gate-nova-docs http://docs-draft.openstack.org/12/209812/2/check/gate-nova-docs/20f05dd//doc/build/html/ : SUCCESS in 15m 19s\n- gate-nova-python27 http://logs.openstack.org/12/209812/2/check/gate-nova-python27/73faf19/ : SUCCESS in 10m 41s\n- gate-nova-python34 http://logs.openstack.org/12/209812/2/check/gate-nova-python34/cf30964/ : SUCCESS in 10m 10s\n- gate-nova-requirements http://logs.openstack.org/12/209812/2/check/gate-nova-requirements/aaae430/ : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 56s\n- gate-tempest-dsvm-full http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-full/d1c75b6/ : SUCCESS in 50m 53s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-postgres-full/1dbcb10/ : SUCCESS in 50m 13s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-neutron-full/9c336d2/ : SUCCESS in 1h 04m 35s\n- gate-grenade-dsvm http://logs.openstack.org/12/209812/2/check/gate-grenade-dsvm/0649764/ : SUCCESS in 42m 35s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-large-ops/44589e8/ : SUCCESS in 38m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-neutron-large-ops/db98558/ : SUCCESS in 24m 34s\n- gate-devstack-dsvm-cells http://logs.openstack.org/12/209812/2/check/gate-devstack-dsvm-cells/633ff94/ : SUCCESS in 28m 38s\n- gate-nova-tox-functional http://logs.openstack.org/12/209812/2/check/gate-nova-tox-functional/7a77b04/ : SUCCESS in 10m 17s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/12/209812/2/check/gate-grenade-dsvm-partial-ncpu/23d79c5/ : SUCCESS in 46m 37s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-ironic-pxe_ssh/544a0ff/ : SUCCESS in 34m 49s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-nova-v21-full/77e92bd/ : SUCCESS in 50m 28s\n- gate-tempest-dsvm-cells http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-cells/883c060/ : SUCCESS in 38m 33s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-full-ceph/c134ae7/ : SUCCESS in 43m 15s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-multinode-full/d39fc02/ : SUCCESS in 1h 05m 10s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/12/209812/2/check/gate-tempest-dsvm-neutron-multinode-full/055d7f2/ : SUCCESS in 1h 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74aaa4b9588c18e98da1b80234818cab875e7f4c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-08-06 11:40:32.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/209812/2/ : FAILURE in 1h 49m 43s","accounts_in_message":[],"_revision_number":2},{"id":"19cfb4ba812856af3ad08ee2fdbd4231d6021292","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-08-07 01:05:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/209812/2/Nova-ML2-Sriov : SUCCESS in 1h 39m 50s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"41cd19fbe9dc6c881fefba2e29977d9de829b8a4","author":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"date":"2015-08-08 12:20:08.000000000","message":"Abandoned\n\nthis is wrong patch. coreect way is upgrade oslo.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7a2aa9d8b8e28b4821f6872b39783a4181d9b66","revisions":{"42157a07fd14ee28a60ce81604db42e714f462d4":{"kind":"REWORK","_number":1,"created":"2015-08-06 08:17:27.000000000","uploader":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"ref":"refs/changes/12/209812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/209812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/209812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/209812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/209812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/209812/1"}}},"commit":{"parents":[{"commit":"70165ed7dc0b1e1fa49f55861093938375942992","subject":"Merge \"allow live migration in case of a booted from volume instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70165ed7dc0b1e1fa49f55861093938375942992"}]}],"author":{"name":"Yongli He","email":"yongli.he@intel.com","date":"2015-08-06 08:10:57.000000000","tz":480},"committer":{"name":"Yongli He","email":"yongli.he@intel.com","date":"2015-08-06 08:16:37.000000000","tz":480},"subject":"make PciDevice and VirtualInterface compatable with dict","message":"make PciDevice and VirtualInterface compatable with dict\n\nPciDevice VirtualInterface object need dict like behevia to\nwork with new oslo_versionedobjectst.\n\nChange-Id: I53d9c327cb902dc150eac3b9b672d4a6f89c7fcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/42157a07fd14ee28a60ce81604db42e714f462d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/42157a07fd14ee28a60ce81604db42e714f462d4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"70165ed7dc0b1e1fa49f55861093938375942992","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d7a2aa9d8b8e28b4821f6872b39783a4181d9b66":{"kind":"REWORK","_number":2,"created":"2015-08-06 09:08:45.000000000","uploader":{"_account_id":7543,"name":"Yongli He","email":"yongli.he@intel.com","username":"yongli.he"},"ref":"refs/changes/12/209812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/209812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/209812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/209812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/209812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/209812/2"}}},"commit":{"parents":[{"commit":"70165ed7dc0b1e1fa49f55861093938375942992","subject":"Merge \"allow live migration in case of a booted from volume instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/70165ed7dc0b1e1fa49f55861093938375942992"}]}],"author":{"name":"Yongli He","email":"yongli.he@intel.com","date":"2015-08-06 08:10:57.000000000","tz":480},"committer":{"name":"Yongli He","email":"yongli.he@intel.com","date":"2015-08-06 09:06:11.000000000","tz":480},"subject":"use new oslo.versionedobjects","message":"use new oslo.versionedobjects\n\nDict compact is removed but this need oslo.versionedobjects\u003e\u003d0.6.0\n\nChange-Id: I53d9c327cb902dc150eac3b9b672d4a6f89c7fcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d7a2aa9d8b8e28b4821f6872b39783a4181d9b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d7a2aa9d8b8e28b4821f6872b39783a4181d9b66"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"70165ed7dc0b1e1fa49f55861093938375942992","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
