)]}'
{"id":"openstack%2Fnova~211027","triplet_id":"openstack%2Fnova~master~I02b04eada656b0fd9dd7bab072e10319f83ad291","project":"openstack/nova","branch":"master","topic":"bug/1482029","hashtags":[],"change_id":"I02b04eada656b0fd9dd7bab072e10319f83ad291","subject":"fix bug: Delete \"vm that has snapshot\" spend more time We are now using ceph as nova backend in version juno/stable. In the following scene, deleting instance will take more time than  icehouse.","status":"ABANDONED","created":"2015-08-10 08:10:50.000000000","updated":"2015-09-15 02:45:24.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce962c2923971b68e5cca2ada36cd702ea1b0237","_number":211027,"virtual_id_number":211027,"owner":{"_account_id":9654,"name":"geml","email":"gemiaolong@unitedstack.com","username":"geml"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2015-08-13 00:11:28.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":1,"date":"2015-08-10 09:07:53.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2015-08-10 08:19:27.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-08-10 09:19:19.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2015-08-10 08:11:00.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":-1,"date":"2015-08-10 09:36:02.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"date":"2015-08-10 08:46:06.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2015-08-10 08:26:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-08-10 17:10:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-10 08:11:00.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2015-08-10 08:19:27.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-10 08:26:57.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-10 08:46:06.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-10 09:07:53.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-10 09:19:19.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-08-10 09:36:02.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-08-10 17:10:13.000000000","updated_by":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"reviewer":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"state":"REVIEWER"},{"updated":"2015-08-13 00:11:28.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58095b1fe9bd21d42b9345a0298456bf9156a60f","author":{"_account_id":9654,"name":"geml","email":"gemiaolong@unitedstack.com","username":"geml"},"date":"2015-08-10 08:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26237cf8320a19831822bdf5f08eafc6068b07e7","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2015-08-10 08:11:00.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. Also it\n    takes generally at least a couple of weeks for cores to get around to\n    reviewing code. The more you participate in the community the more\n    rewarding it is for you. You may also notice that the more you get to know\n    people and get to be known, the faster your patches will be reviewed and\n    eventually approved. Get to know others and become known by doing code\n    reviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: http://docs.openstack.org/infra/manual/developers.html\n    Commit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"b93fbbb5e1fabc63b4f323aee8c36724b3677a24","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-08-10 08:19:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- hyper-v-ci http://64.119.130.115/211027/1/ : FAILURE in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c18e5569615292dfb976f1aefcc10f14feb2773a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-08-10 08:26:57.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/27/211027/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4b6f6cea05f84930b3cd400480d14b5c/index.html : SUCCESS in 13m 59s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/27/211027/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8df815508e7145ac9ba13790db746a5f/index.html : SUCCESS in 13m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/27/211027/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/11e90e4c33a14730a9851bce465271c0/index.html : SUCCESS in 14m 04s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/27/211027/1/check/real-db-upgrade_nova_percona_user_001:th-percona/e348c60bd8be4556a8bea9bb6c20e56f/index.html : SUCCESS in 15m 06s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/27/211027/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e884a98b8daf4b9eac5c4791fc628a35/index.html : SUCCESS in 11m 36s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/27/211027/1/check/real-db-upgrade_nova_percona_user_002:th-percona/32d72d3bb26442179bce0eb12111cd67/index.html : SUCCESS in 12m 01s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"2cdc81cd569f158dcd7440265062e2a828eb78a9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-08-10 08:46:06.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/211027/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"df61ab86b03cb61c6aa8063633b575f7e25dbe8c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-08-10 09:07:53.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/27/211027/1/7639/index.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":1},{"id":"6999469db42f0d20abd1193848d283cb287f133b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-10 09:19:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/27/211027/1/check/gate-nova-pep8/0cd849f/ : SUCCESS in 6m 42s\n- gate-nova-docs http://docs-draft.openstack.org/27/211027/1/check/gate-nova-docs/98e27b8//doc/build/html/ : SUCCESS in 9m 19s\n- gate-nova-python27 http://logs.openstack.org/27/211027/1/check/gate-nova-python27/0a50c54/ : FAILURE in 10m 16s\n- gate-nova-python34 http://logs.openstack.org/27/211027/1/check/gate-nova-python34/68962cf/ : SUCCESS in 17m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-full/cbf8721/ : SUCCESS in 48m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-postgres-full/738ecba/ : SUCCESS in 43m 38s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-neutron-full/b50a986/ : SUCCESS in 51m 52s\n- gate-grenade-dsvm http://logs.openstack.org/27/211027/1/check/gate-grenade-dsvm/5136c31/ : SUCCESS in 51m 09s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-large-ops/1849fc5/ : SUCCESS in 27m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-neutron-large-ops/d0445e4/ : SUCCESS in 31m 30s\n- gate-devstack-dsvm-cells http://logs.openstack.org/27/211027/1/check/gate-devstack-dsvm-cells/25506cd/ : SUCCESS in 35m 00s\n- gate-nova-tox-functional http://logs.openstack.org/27/211027/1/check/gate-nova-tox-functional/f04b88a/ : SUCCESS in 13m 41s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/27/211027/1/check/gate-grenade-dsvm-partial-ncpu/af0c410/ : SUCCESS in 48m 37s\n- gate-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-ironic-pxe_ssh/53023d8/ : SUCCESS in 44m 53s\n- gate-tempest-dsvm-nova-v21-full http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-nova-v21-full/45467ff/ : SUCCESS in 57m 02s\n- gate-tempest-dsvm-cells http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-cells/ba1e5b4/ : SUCCESS in 45m 08s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-full-ceph/574c009/ : SUCCESS in 51m 33s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-multinode-full/67ad053/ : SUCCESS in 52m 29s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/27/211027/1/check/gate-tempest-dsvm-neutron-multinode-full/aece2d6/ : SUCCESS in 1h 03m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb204205790b25a8cf1dab0121a389556bb87991","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-08-10 09:36:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/27/211027/1/check/dsvm-tempest-xen/d92974a : FAILURE in 1h 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1d2f3732db8d4cdf178481370465442074471213","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2015-08-10 17:10:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fa04801f0b3a61566dc79c440aa1c5a4c19d513c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-08-13 00:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/211027/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/211027/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4607b486330b67df78e191721578c3c3a55405fb","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2015-09-15 02:45:24.000000000","message":"Abandoned\n\nThis patch has been stalled for quite a while, so I am going to abandon it to keep the code review queue sane. Please restore the change when it is ready for review.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19f870243b6757e7d8b5b41e638c943d82054a20","revisions":{"19f870243b6757e7d8b5b41e638c943d82054a20":{"kind":"REWORK","_number":1,"created":"2015-08-10 08:10:50.000000000","uploader":{"_account_id":9654,"name":"geml","email":"gemiaolong@unitedstack.com","username":"geml"},"ref":"refs/changes/27/211027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/27/211027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/27/211027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/27/211027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/27/211027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/27/211027/1"}}},"commit":{"parents":[{"commit":"f091ff04e409cfc968d7ca38bdc5016e0de84eef","subject":"Merge \"docs: add link to liberty summit session on v2.1 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f091ff04e409cfc968d7ca38bdc5016e0de84eef"}]}],"author":{"name":"geml","email":"gemiaolong@unitedstack.com","date":"2015-08-10 07:53:35.000000000","tz":480},"committer":{"name":"geml","email":"gemiaolong@unitedstack.com","date":"2015-08-10 08:08:38.000000000","tz":480},"subject":"fix bug: Delete \"vm that has snapshot\" spend more time We are now using ceph as nova backend in version juno/stable. In the following scene, deleting instance will take more time than  icehouse.","message":"fix bug: Delete \"vm that has snapshot\" spend more time\nWe are now using ceph as nova backend in version juno/stable.\nIn the following scene, deleting instance will take more time than\n icehouse.\n\n1. Create an instance.\n2. Wait for instance becoming \"Active\" status.\n3. Wait for instance having fixed ip address.\n4. Create an snapshot from the new instance.\n5. Wait for image snapshot becoming \"Active\" status.\n6. Delete the instance created in step 1.\n\nin icehouse, we can delete instance in 10 seconds. Recently we upg\n-rade to juno, the time increased to 30~40 seconds which was disco\n-vered by our Monitoring System.\n\nRelated-Bug: #1482029\nChange-Id: I02b04eada656b0fd9dd7bab072e10319f83ad291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/19f870243b6757e7d8b5b41e638c943d82054a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/19f870243b6757e7d8b5b41e638c943d82054a20"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f091ff04e409cfc968d7ca38bdc5016e0de84eef","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
