)]}'
{"id":"openstack%2Fnova~23207","triplet_id":"openstack%2Fnova~stable%2Ffolsom~I72966fa205973ec638816b07bfdcd54f1102feb5","project":"openstack/nova","branch":"stable/folsom","topic":"bug/1133495","hashtags":[],"change_id":"I72966fa205973ec638816b07bfdcd54f1102feb5","subject":"Shrink size of aggregate_metadata_get_by_host sql query","status":"MERGED","created":"2013-02-28 20:53:52.000000000","updated":"2013-05-31 13:48:34.000000000","submitted":"2013-05-31 13:48:31.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16e0192c6daa13cba0f5fbad12f55117e3309539","_number":23207,"virtual_id_number":23207,"owner":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-05-31 13:48:31.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-05-06 23:44:38.000000000","_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":2,"date":"2013-05-31 13:03:31.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":1,"date":"2013-05-15 17:35:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-05-06 23:44:38.000000000","_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":1,"date":"2013-05-31 13:03:31.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-03-04 23:58:48.000000000","updated_by":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"reviewer":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"state":"REVIEWER"},{"updated":"2013-03-19 19:32:21.000000000","updated_by":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"reviewer":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"state":"REVIEWER"},{"updated":"2013-05-06 23:44:38.000000000","updated_by":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"reviewer":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"state":"REVIEWER"},{"updated":"2013-05-15 17:35:27.000000000","updated_by":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"reviewer":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"state":"REVIEWER"},{"updated":"2013-05-31 13:03:31.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2013-05-31 13:48:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"877558e2a1a05ef0e01f34a009c7e209990e9373","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-02-28 21:20:48.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/1/check/gate-nova-pep8/21829 : SUCCESS\n- http://logs.openstack.org/23207/1/check/gate-nova-docs/16317 : SUCCESS\n- http://logs.openstack.org/23207/1/check/gate-nova-python26/17745 : FAILURE\n- http://logs.openstack.org/23207/1/check/gate-nova-python27/17815 : FAILURE\n- http://logs.openstack.org/23207/1/check/gate-tempest-devstack-vm-smoke/156 : SUCCESS\n- http://logs.openstack.org/23207/1/check/gate-tempest-devstack-vm-cinder/6015 : SUCCESS\n- http://logs.openstack.org/23207/1/check/gate-nova-pylint/8966 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"29fa447b6b3c73afaa6a2c36105942501027bac9","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-02-28 23:00:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30c6de08c0bef24bfe3bdc32e835ffdf905aad3e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-02-28 23:27:47.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/2/check/gate-nova-pep8/21844 : SUCCESS\n- http://logs.openstack.org/23207/2/check/gate-nova-docs/16326 : SUCCESS\n- http://logs.openstack.org/23207/2/check/gate-nova-python26/17754 : FAILURE\n- http://logs.openstack.org/23207/2/check/gate-nova-python27/17824 : FAILURE\n- http://logs.openstack.org/23207/2/check/gate-tempest-devstack-vm-smoke/157 : FAILURE\n- http://logs.openstack.org/23207/2/check/gate-tempest-devstack-vm-cinder/6016 : FAILURE\n- http://logs.openstack.org/23207/2/check/gate-nova-pylint/8973 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"9a07035177133e67e28e428247ad7b74be86b584","author":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"date":"2013-03-01 02:38:17.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\nlooks like there is a missing joinedload somewhere in folsom.","accounts_in_message":[],"_revision_number":2},{"id":"d0543a9df8f8e8b95ce12f0b919f5b7efafe8438","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-03-01 23:46:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1fa36785d323e8d2a7723bb6f6f762d0880aaa13","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-03-02 00:09:47.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/3/check/gate-nova-pep8/22015 : FAILURE\n- gate-nova-docs : SKIPPED\n- gate-nova-python26 : SKIPPED\n- gate-nova-python27 : SKIPPED\n- gate-tempest-devstack-vm-smoke : SKIPPED\n- gate-tempest-devstack-vm-cinder : SKIPPED\n- gate-nova-pylint : SKIPPED (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"2a50581a268cba045cf21514e515c42a8ec004fc","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-03-02 00:24:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bcc9f4c4620b86dde5ebe1fe6b0919850cf54378","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-03-02 00:56:59.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/4/check/gate-nova-pep8/22030 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-docs/16500 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-python26/17917 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-python27/17997 : FAILURE\n- http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-smoke/163 : FAILURE\n- http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-cinder/6022 : FAILURE\n- http://logs.openstack.org/23207/4/check/gate-nova-pylint/9038 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"fdedc003f36674675c3045e248f82db7ccb07f45","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-03-02 01:59:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"60d2988d5b62e008823f30772524849e48eb32d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-03-02 02:36:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/4/check/gate-nova-pep8/22071 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-docs/16541 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-python26/17956 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-python27/18038 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-smoke/176 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-cinder/6035 : FAILURE\n- http://logs.openstack.org/23207/4/check/gate-nova-pylint/9040 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"d9db006139ca113a28073b02667ed112d0274e8e","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-03-02 19:55:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"61ef8cd39b3fd072b1954c58fb4b71baadfb7150","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-03-02 20:20:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/4/check/gate-nova-pep8/22106 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-docs/16573 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-python26/17988 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-python27/18070 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-smoke/184 : FAILURE\n- http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-cinder/6043 : SUCCESS\n- http://logs.openstack.org/23207/4/check/gate-nova-pylint/9044 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"6689b468a69b1a58914115ed20eee9e42f3102f8","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-03-04 19:37:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b43ee9ad2624ebe4c61c9f6a29339d694c383e8a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-03-04 20:05:34.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/23207/5/check/gate-nova-pep8/22149 : SUCCESS\n- http://logs.openstack.org/23207/5/check/gate-nova-docs/16613 : SUCCESS\n- http://logs.openstack.org/23207/5/check/gate-nova-python26/18028 : SUCCESS\n- http://logs.openstack.org/23207/5/check/gate-nova-python27/18110 : SUCCESS\n- http://logs.openstack.org/23207/5/check/gate-tempest-devstack-vm-smoke/188 : SUCCESS\n- http://logs.openstack.org/23207/5/check/gate-tempest-devstack-vm-cinder/6047 : SUCCESS\n- http://logs.openstack.org/23207/5/check/gate-nova-pylint/9069 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"ae3165af11dde6e935fcdb9f59e7fb3c30e7c414","author":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"date":"2013-03-04 20:51:31.000000000","message":"Patch Set 5: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"0317420cc09e827d423f9e78012dfa5f66999919","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-03-04 23:58:48.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"0fe4c7f2c0e0372aedffd4c072dbeb44d5bc196a","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-03-19 06:03:02.000000000","message":"Patch Set 5:\n\ncode review expired after 2 weeks of no activity, it can be restored using the `Restore Change` button  under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":5},{"id":"8ea2a40b8608aefce76e838121a861f048af3730","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-03-19 06:03:02.000000000","message":"Patch Set 5: Abandoned\n\ncode review expired after 2 weeks of no activity, it can be restored using the `Restore Change` button  under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":5},{"id":"6f1b6127576437ad6551f089041b15d5cf187c29","author":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"date":"2013-03-19 16:15:53.000000000","message":"Patch Set 5:\n\nplease revive this.","accounts_in_message":[],"_revision_number":5},{"id":"043961a1e7c235807d338d4c6330493c3936611a","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-03-19 17:52:06.000000000","message":"Patch Set 5: Restored","accounts_in_message":[],"_revision_number":5},{"id":"c96033121c11ecd78666f1ab433dc18edeedd55d","author":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"date":"2013-03-19 19:32:21.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"a1e9e19e75ec69bef98a9f13f81e88001075e9f1","author":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"date":"2013-03-19 19:39:02.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"84c3acf02269b07710a9ccf2becd759d6b0f322f","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2013-03-26 07:05:48.000000000","message":"Patch Set 5: I would prefer that you didn\u0027t merge this\n\nThanks for the backport Joe\n\nI\u0027m worried about the implications of the change to _aggregate_get_query()\n\nThe change comes from I4048ea6f477f7a695701eb3ba57f97afca0c0e41 and you can see how Russell deletes the hosts/metadata keys from the the aggregate in nova.compute.api for API compat reasons\n\nI can\u0027t see I fully grok yet Russell\u0027s changes, but neither do I fully understand why they needed to be backported along with this patch\n\nCare to give more details?","accounts_in_message":[],"_revision_number":5},{"id":"7c149f6a91fd4bb20e0fd14a8f656f009d9375eb","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-04-03 06:03:06.000000000","message":"Patch Set 5:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":5},{"id":"e1450932d35752a0a546b065e9e12d5c972ec132","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-04-03 06:03:06.000000000","message":"Patch Set 5: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":5},{"id":"79e27172601d9fcb2a3c033935ec6096837bd621","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-04-03 19:41:50.000000000","message":"Patch Set 5:\n\nMark, Good catch\n\nRussell\u0027s change was needed here to prevent some sqlalchemy lazy load issues (see req-d442441d-3eca-46c8-8d09-d1344c3b7630 in http://logs.openstack.org/23207/4/check/gate-tempest-devstack-vm-smoke/184/logs/screen-n-api.txt.gz).  As for only backporting part of the patch, I think you are right I need to backport the entire patch.  I will verify if the host/metadata del piece is needed here.","accounts_in_message":[],"_revision_number":5},{"id":"ad42d01838352b8a437c64a90cd9c3027b73fd41","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-04-03 20:26:33.000000000","message":"Patch Set 5: Restored","accounts_in_message":[],"_revision_number":5},{"id":"5eeb30b382c281cd6ef332a1a407caf24b7fff65","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-04-03 20:26:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d08d3486c8929cbe71ec7e1a9a35cb0c445f5131","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-04-03 20:47:52.000000000","message":"Patch Set 6: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/23207/6/check/gate-nova-pep8/24052 : SUCCESS\n- http://logs.openstack.org/23207/6/check/gate-nova-docs/18315 : SUCCESS\n- http://logs.openstack.org/23207/6/check/gate-nova-python26/19723 : FAILURE\n- http://logs.openstack.org/23207/6/check/gate-nova-python26-rhel6/422 : FAILURE (non-voting)\n- http://logs.openstack.org/23207/6/check/gate-nova-python27/19811 : FAILURE\n- http://logs.openstack.org/23207/6/check/gate-tempest-devstack-vm-smoke/484 : SUCCESS\n- http://logs.openstack.org/23207/6/check/gate-tempest-devstack-vm-cinder/6311 : SUCCESS\n- http://logs.openstack.org/23207/6/check/gate-nova-pylint/10126 : SUCCESS (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"73767ff6fae85aa1dbaa9d811f780987f66bdcd9","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-04-03 22:20:12.000000000","message":"Patch Set 6: Abandoned\n\nBecause this only a performance fix and its touching sqlalchemy code, which can fail in unusual ways, this may not be worth back porting.","accounts_in_message":[],"_revision_number":6},{"id":"de80f75038d86fdacda796d0c1f0a4e1743b4e5f","author":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"date":"2013-04-04 16:40:04.000000000","message":"Patch Set 6:\n\nI think this is backport worthy:\n\n\"When I try to do the same in our production environment which has 80 compute nodes (splitting them again into 2 aggregates) it doesn\u0027t work.\n\nnova-scheduler starts to go very slow, I scheduled an instance and gave up after 5 minutes, it seemed to be taking ages and the host was at 100% cpu. Also got about 500 messages in rabbit that were unacknowledged.\"\n\nBasically aggregates are completely unusable in production without this.","accounts_in_message":[],"_revision_number":6},{"id":"b991b962f4678c45b7ad96c746d45ce47bf3cf81","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-04-30 22:05:06.000000000","message":"Patch Set 6: Restored\n\nWIP","accounts_in_message":[],"_revision_number":6},{"id":"91b3dc8fb9337a9027d060d9e9cd1e6316257f35","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-04-30 22:05:10.000000000","message":"Patch Set 6: Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"f2c935dd9c869452e96b92e50ef39ba5f10b7b95","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-05-01 18:47:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"10cbf91e418a54befea971c507877d291b25f51e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-01 19:21:48.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/23207/7/check/gate-nova-pep8/25183 : SUCCESS in 00:02:11\n- http://docs-draft.openstack.org/23207/7/check/gate-nova-docs/19401/doc/build/html/ : SUCCESS in 00:07:02\n- http://logs.openstack.org/23207/7/check/gate-nova-python26/20809 : SUCCESS in 00:18:32\n- http://logs.openstack.org/23207/7/check/gate-nova-python26-centos6/67 : FAILURE in 00:17:13 (non-voting)\n- http://logs.openstack.org/23207/7/check/gate-nova-python27/20897 : SUCCESS in 00:12:46\n- http://logs.openstack.org/23207/7/check/gate-tempest-devstack-vm-smoke/546 : SUCCESS in 00:11:03\n- http://logs.openstack.org/23207/7/check/gate-tempest-devstack-vm-cinder/6367 : SUCCESS in 00:13:05\n- http://logs.openstack.org/23207/7/check/gate-nova-pylint/10953 : SUCCESS in 00:07:00 (non-voting)\n","accounts_in_message":[],"_revision_number":7},{"id":"76e79ce81daf068ea0255fdf796f8c1469b57122","author":{"_account_id":67,"name":"Vish Ishaya","email":"vishvananda@gmail.com","username":"vishvananda"},"date":"2013-05-06 23:44:38.000000000","message":"Patch Set 7: Looks good to me (core reviewer)\n\nstill lgtm","accounts_in_message":[],"_revision_number":7},{"id":"463eb05a588aab3bf173a004e65093d256f45c15","author":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"date":"2013-05-15 17:35:27.000000000","message":"Patch Set 7: Looks good to me, but someone else must approve\n\nlgtm","accounts_in_message":[],"_revision_number":7},{"id":"c577be4c5cf7a1ea1c872dc668b4fdfbf78c901b","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-05-30 06:03:02.000000000","message":"Patch Set 7:\n\ncode review expired after 2 weeks of no activity, it can be restored using the `Restore Change` button  under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":7},{"id":"0551935b0afde8207f645638693436c8e045b811","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-05-30 06:03:02.000000000","message":"Patch Set 7: Abandoned\n\ncode review expired after 2 weeks of no activity, it can be restored using the `Restore Change` button  under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":7},{"id":"1dac1974e6f8c2901035d89c869e7ca498820362","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-05-30 21:10:38.000000000","message":"Patch Set 7: Restored","accounts_in_message":[],"_revision_number":7},{"id":"224d62b677f17a19a46e6a11ec2eb98978f3f4b4","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2013-05-31 13:03:31.000000000","message":"Patch Set 7: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":7},{"id":"606e59e169a3c190e1f547dab57b149cd145f80b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-31 13:03:52.000000000","message":"Patch Set 7: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"3bb937a6606b293b3a260f34505139d433583d4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-31 13:48:31.000000000","message":"Patch Set 7: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/23207/7/gate/gate-nova-docs/21670/doc/build/html/ : SUCCESS in 9m 57s\n- http://logs.openstack.org/23207/7/gate/gate-nova-pep8/27450 : SUCCESS in 2m 23s\n- http://logs.openstack.org/23207/7/gate/gate-nova-python27/23160 : SUCCESS in 13m 12s\n- http://logs.openstack.org/23207/7/gate/gate-tempest-devstack-vm-smoke/599 : SUCCESS in 13m 19s\n- http://logs.openstack.org/23207/7/gate/gate-tempest-devstack-vm-cinder/6420 : SUCCESS in 12m 58s\n","accounts_in_message":[],"_revision_number":7},{"id":"cccc1fe97253274cf580f0f066dc456b06e39cbf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-31 13:48:34.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e11d90af9c30d9af875974520cef207a839a0e21","revisions":{"807b538e9d8f33711310c3013623ca0a24ef9b26":{"kind":"REWORK","_number":1,"created":"2013-02-28 20:53:52.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/1"}}},"commit":{"parents":[{"commit":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","subject":"libvirt: Fix nova-compute start when missing ip.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5e7f5512435fe0ca264be28ef23fe2ebb449d1c"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-27 20:59:08.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-28 19:42:48.000000000","tz":-480},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/807b538e9d8f33711310c3013623ca0a24ef9b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/807b538e9d8f33711310c3013623ca0a24ef9b26"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","is_merged_in_target_branch":true,"change_id":"Id383544b44e64205fc3b4f850d0d11ad2ebd5da7","change_number":22583,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"},"7c7ecf14f73117429c748b6c221146641e56ffe0":{"kind":"REWORK","_number":2,"created":"2013-02-28 23:00:28.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/2"}}},"commit":{"parents":[{"commit":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","subject":"libvirt: Fix nova-compute start when missing ip.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5e7f5512435fe0ca264be28ef23fe2ebb449d1c"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-27 20:59:08.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-28 22:56:35.000000000","tz":-480},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c7ecf14f73117429c748b6c221146641e56ffe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c7ecf14f73117429c748b6c221146641e56ffe0"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","is_merged_in_target_branch":true,"change_id":"Id383544b44e64205fc3b4f850d0d11ad2ebd5da7","change_number":22583,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"},"16d2a304f2539017f73d66e3bb26187a7a63bdd8":{"kind":"REWORK","_number":3,"created":"2013-03-01 23:46:24.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/3"}}},"commit":{"parents":[{"commit":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","subject":"libvirt: Fix nova-compute start when missing ip.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5e7f5512435fe0ca264be28ef23fe2ebb449d1c"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-27 20:59:08.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-03-01 23:45:54.000000000","tz":-480},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16d2a304f2539017f73d66e3bb26187a7a63bdd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/16d2a304f2539017f73d66e3bb26187a7a63bdd8"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","is_merged_in_target_branch":true,"change_id":"Id383544b44e64205fc3b4f850d0d11ad2ebd5da7","change_number":22583,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"},"8e8af5c7ce8680ea42e5026b235bcf52f4c9d5db":{"kind":"REWORK","_number":4,"created":"2013-03-02 00:24:19.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/4"}}},"commit":{"parents":[{"commit":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","subject":"libvirt: Fix nova-compute start when missing ip.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5e7f5512435fe0ca264be28ef23fe2ebb449d1c"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-27 20:59:08.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-03-02 00:23:52.000000000","tz":-480},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e8af5c7ce8680ea42e5026b235bcf52f4c9d5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e8af5c7ce8680ea42e5026b235bcf52f4c9d5db"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","is_merged_in_target_branch":true,"change_id":"Id383544b44e64205fc3b4f850d0d11ad2ebd5da7","change_number":22583,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"},"4c0c1c7ea4b2cb47eaac5c26d849349961d6f211":{"kind":"REWORK","_number":5,"created":"2013-03-04 19:37:25.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/5"}}},"commit":{"parents":[{"commit":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","subject":"libvirt: Fix nova-compute start when missing ip.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5e7f5512435fe0ca264be28ef23fe2ebb449d1c"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-02-27 20:59:08.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-03-04 19:37:08.000000000","tz":-480},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c0c1c7ea4b2cb47eaac5c26d849349961d6f211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c0c1c7ea4b2cb47eaac5c26d849349961d6f211"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","is_merged_in_target_branch":true,"change_id":"Id383544b44e64205fc3b4f850d0d11ad2ebd5da7","change_number":22583,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"},"c98b896ee85ad76ac1c297f52c40f1e40e3ee6a1":{"kind":"REWORK","_number":6,"created":"2013-04-03 20:26:54.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/6"}}},"commit":{"parents":[{"commit":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","subject":"libvirt: Fix nova-compute start when missing ip.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5e7f5512435fe0ca264be28ef23fe2ebb449d1c"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-04-03 20:25:53.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-04-03 20:26:08.000000000","tz":0},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c98b896ee85ad76ac1c297f52c40f1e40e3ee6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c98b896ee85ad76ac1c297f52c40f1e40e3ee6a1"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"d5e7f5512435fe0ca264be28ef23fe2ebb449d1c","is_merged_in_target_branch":true,"change_id":"Id383544b44e64205fc3b4f850d0d11ad2ebd5da7","change_number":22583,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"},"e11d90af9c30d9af875974520cef207a839a0e21":{"kind":"REWORK","_number":7,"created":"2013-05-01 18:47:56.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/07/23207/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/07/23207/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/07/23207/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/07/23207/7"}}},"commit":{"parents":[{"commit":"9ecd965eaf03dac01f81964a1e782223d2e2d1a8","subject":"Bump version to 2012.2.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ecd965eaf03dac01f81964a1e782223d2e2d1a8"}]}],"author":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-04-03 20:25:53.000000000","tz":0},"committer":{"name":"Joe Gordon","email":"jogo@cloudscaling.com","date":"2013-05-01 18:47:28.000000000","tz":0},"subject":"Shrink size of aggregate_metadata_get_by_host sql query","message":"Shrink size of aggregate_metadata_get_by_host sql query\n\n* Remove unnecessary secondary joins in Aggregate model.\n* Remove unnecessary backrefs (only needed to create bidirectional relationship)\n\naggregate_metadata_get_by_host goes from 6 JOINS down to two JOINS.\n\nFix bug 1133495\n\nChange-Id: I72966fa205973ec638816b07bfdcd54f1102feb5\n(cherry picked from commit ce23d6c1d679355bafede816bca39ea1eb24073c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e11d90af9c30d9af875974520cef207a839a0e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e11d90af9c30d9af875974520cef207a839a0e21"}]},"parents_data":[{"branch_name":"refs/heads/stable/folsom","commit_id":"9ecd965eaf03dac01f81964a1e782223d2e2d1a8","is_merged_in_target_branch":true,"change_id":"Ibf237525fc86d5ba2129620f58d574685a713782","change_number":26789,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/folsom"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
