)]}'
{"id":"openstack%2Fnova~234090","triplet_id":"openstack%2Fnova~master~I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748","project":"openstack/nova","branch":"master","topic":"live-migration-to-liveMigration","hashtags":[],"change_id":"I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748","subject":"Trivial fix to use constant instead of string for instance action","status":"ABANDONED","created":"2015-10-13 09:55:21.000000000","updated":"2015-11-27 18:03:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7cac11b72a3c22a16140c5e42723dc62b2c0076","_number":234090,"virtual_id_number":234090,"owner":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},{"value":1,"date":"2015-10-19 13:36:14.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2015-10-19 13:36:53.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2015-10-21 14:06:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":1,"date":"2015-10-19 16:18:04.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2015-11-08 17:15:03.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"date":"2015-10-21 11:43:20.000000000","_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},{"date":"2015-10-20 01:42:22.000000000","_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"disliked":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":-1,"date":"2015-10-19 15:52:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},{"value":1,"date":"2015-10-19 13:17:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-10-13 15:53:45.000000000","updated_by":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"reviewer":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"state":"REVIEWER"},{"updated":"2015-10-13 20:41:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-14 06:08:22.000000000","updated_by":{"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},"reviewer":{"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},"state":"REVIEWER"},{"updated":"2015-10-14 06:51:51.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2015-10-15 05:01:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-15 05:15:12.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-15 05:46:18.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-15 16:13:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-19 13:17:27.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2015-10-19 13:36:14.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-10-19 13:36:53.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-19 15:52:48.000000000","updated_by":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"reviewer":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"state":"REVIEWER"},{"updated":"2015-10-19 16:18:04.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-10-21 11:43:20.000000000","updated_by":{"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},"reviewer":{"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},"state":"REVIEWER"},{"updated":"2015-10-21 14:06:07.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-11-08 17:15:03.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a04b4d9c61d5da399ce725f0de81b75c7b885970","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-13 09:55:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be9c8adca91d6f18c32bb38930d0bc221fd1da7f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-13 10:56:11.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/90/234090/1/13510/index.html : SUCCESS in 56 m 10 s","accounts_in_message":[],"_revision_number":1},{"id":"14353e1baded9bb31c518a536692f4466ce7c901","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-13 11:06:32.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/27d81a51f1ad4ce3acf9ec4335673edb/index.html : SUCCESS in 15m 15s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3ac232a6b38f44cd880ac2c8e5bf1b00/index.html : SUCCESS in 15m 45s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d7e1989d377944e680bdc7a6a1b6068a/index.html : SUCCESS in 11m 43s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/1/check/real-db-upgrade_nova_percona_user_001:th-percona/370ae8f1e9674b20ac6b8c8ae2da1823/index.html : SUCCESS in 15m 15s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/69eb97929b4b4c688a054ee75f685471/index.html : SUCCESS in 12m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/1/check/real-db-upgrade_nova_percona_user_002:th-percona/709d45ed282b477a99c0ded5f0db65a6/index.html : SUCCESS in 11m 43s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"6fcf1206ba93644f4149d8e0a113c670dd4f366a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-13 11:10:47.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/90/234090/1/check/check-ibm-tempest-dsvm-full/5acac4c/ : SUCCESS in 50m 34s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/90/234090/1/check/check-ibm-tempest-dsvm-postgres-full/b34df27/ : SUCCESS in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7f5cc41cf00ec36eb83ae2f446d42cf5f3349d9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-13 11:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/90/234090/1/check/dsvm-tempest-xen/0aa8491 : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b21de794dff99fda0b6678c909d9420ce34a98ed","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-13 12:37:03.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/1/ : FAILURE in 2h 23m 11s","accounts_in_message":[],"_revision_number":1},{"id":"36ddea0eec3e5a38c137f1ec2d850412c0d7af90","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-13 15:40:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/90/234090/1/check/gate-nova-pep8/f941e9d/ : SUCCESS in 8m 21s\n- gate-nova-docs http://docs-draft.openstack.org/90/234090/1/check/gate-nova-docs/7e48ffe//doc/build/html/ : SUCCESS in 14m 26s\n- gate-nova-python27 http://logs.openstack.org/90/234090/1/check/gate-nova-python27/1179822/ : FAILURE in 16m 22s\n- gate-nova-python34 http://logs.openstack.org/90/234090/1/check/gate-nova-python34/66990d8/ : SUCCESS in 10m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-full/0e3c919/ : SUCCESS in 42m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-postgres-full/53b8d41/ : SUCCESS in 39m 56s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-neutron-full/185258d/ : SUCCESS in 52m 51s\n- gate-grenade-dsvm http://logs.openstack.org/90/234090/1/check/gate-grenade-dsvm/1091fdf/ : SUCCESS in 44m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-large-ops/eff1a52/ : SUCCESS in 28m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-neutron-large-ops/bdd7e37/ : SUCCESS in 26m 16s\n- gate-nova-tox-functional http://logs.openstack.org/90/234090/1/check/gate-nova-tox-functional/f5efe07/ : SUCCESS in 16m 09s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/90/234090/1/check/gate-grenade-dsvm-partial-ncpu/872fa8d/ : SUCCESS in 44m 28s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/26ddb38/ : SUCCESS in 35m 50s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-nova-v20-api/e26331d/ : SUCCESS in 48m 14s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-nova-v20-api-legacy/0bb9ac5/ : SUCCESS in 34m 10s\n- gate-tempest-dsvm-cells http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-cells/aa1c709/ : SUCCESS in 38m 35s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-full-ceph/2920497/ : SUCCESS in 43m 59s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/90/234090/1/check/gate-tempest-dsvm-multinode-full/1267f1c/ : SUCCESS in 1h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54f40b852677f70987139d637e89bb5cae65d2c4","author":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"date":"2015-10-13 15:53:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d5b7da15c9850fb35bf4ad8a129ff78384b9e04","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-13 20:11:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/234090/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/234090/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"08931a6077b71224ce196eea362dff9fd716e19a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-10-13 20:41:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/234090/1/Nova-ML2-Sriov : SUCCESS in 45m 54s (non-voting)\n\nFor more information of Mellanox CI, visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"540c39bdbc100acc575f82e1be7d10fbf1c70418","author":{"_account_id":16066,"name":"xiexs","email":"xiexs@cn.fujitsu.com","username":"simen"},"date":"2015-10-14 06:08:22.000000000","message":"Patch Set 1: Code-Review+1\n\nHi, tang,\n  It LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"0b76fdc324dfb8d14a3d7e607660fc23dfc3aa7f","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2015-10-14 06:51:51.000000000","message":"Patch Set 1: Code-Review-1\n\nIf you want change this, you should also change it in unit test.\nAs this:\nhttp://git.openstack.org/cgit/openstack/nova/tree/nova/tests/unit/compute/test_compute.py#n9888\nthis line records the instance action as before with \u0027live-migration\u0027.\n\nBut in the code:\nhttp://git.openstack.org/cgit/openstack/nova/tree/nova/compute/api.py#n3267\nit records the action as you changed with \u0027liveMigration\u0027\n\nthis might be the reason of your jenkins failure.","accounts_in_message":[],"_revision_number":1},{"id":"730a38c6fec99fdbd46668003cfbbf237cdd234c","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-14 07:23:09.000000000","message":"Patch Set 1:\n\nHi Zheng,\n\nThank you for reviewing. Indeed, I missed it.\n\nWill update it in v2.\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"ecb1bcf055cda145d25328aa6d641d70b9195a71","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-15 04:00:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ba39a9e2555bd2bdb932e43a658c41a73522f74","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-15 04:20:07.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3bf56334282d47c1afce6db383c05363/index.html : SUCCESS in 11m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9d66b55339304fbebdd3b3a158e82c8e/index.html : SUCCESS in 14m 55s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/899f0c995180485abf8f43e14d5ddc3c/index.html : SUCCESS in 14m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/2/check/real-db-upgrade_nova_percona_user_001:th-percona/eec5f9cc71b74261bd492d3a1c1021e7/index.html : SUCCESS in 12m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/ad45ccadbbe74b1e9a88021b70ce2b88/index.html : SUCCESS in 15m 35s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/2/check/real-db-upgrade_nova_percona_user_002:th-percona/e867b2e9d5004466aa2fbc5dc2bef2c3/index.html : SUCCESS in 15m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"fecfb13d1ee75373dd7949cb7ec7af486158beed","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-15 04:29:20.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/234090/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"1e4f2f4d3489c8f44447a4bb0b52490dc6d336b3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-15 04:54:11.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/90/234090/2/13746/index.html : SUCCESS in 50 m 31 s","accounts_in_message":[],"_revision_number":2},{"id":"777ec07c1c29a466ce872714d990f583d3154b69","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-15 05:01:01.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/90/234090/2/check/check-ibm-tempest-dsvm-full/0f900e3/ : SUCCESS in 59m 39s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/90/234090/2/check/check-ibm-tempest-dsvm-postgres-full/0061542/ : SUCCESS in 54m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"279540d3ec3367139d773a1f54ce8a81668176d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-15 05:06:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/234090/2/check/gate-nova-pep8/9fa37fc/ : SUCCESS in 8m 50s\n- gate-nova-docs http://docs-draft.openstack.org/90/234090/2/check/gate-nova-docs/a3d2c95//doc/build/html/ : SUCCESS in 9m 24s\n- gate-nova-python27 http://logs.openstack.org/90/234090/2/check/gate-nova-python27/32da878/ : SUCCESS in 13m 48s\n- gate-nova-python34 http://logs.openstack.org/90/234090/2/check/gate-nova-python34/2990124/ : SUCCESS in 9m 34s\n- gate-tempest-dsvm-full http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-full/947dff2/ : SUCCESS in 50m 13s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-postgres-full/601d0e3/ : SUCCESS in 45m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-neutron-full/db082c8/ : SUCCESS in 1h 05m 41s\n- gate-grenade-dsvm http://logs.openstack.org/90/234090/2/check/gate-grenade-dsvm/9b4df01/ : SUCCESS in 43m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-large-ops/30b55aa/ : SUCCESS in 36m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-neutron-large-ops/885a218/ : SUCCESS in 27m 30s\n- gate-nova-tox-functional http://logs.openstack.org/90/234090/2/check/gate-nova-tox-functional/01f297d/ : SUCCESS in 12m 35s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/90/234090/2/check/gate-grenade-dsvm-partial-ncpu/aa37041/ : SUCCESS in 51m 34s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/1f9d549/ : SUCCESS in 45m 18s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-nova-v20-api/871799f/ : SUCCESS in 31m 31s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-nova-v20-api-legacy/9d44562/ : SUCCESS in 29m 42s\n- gate-tempest-dsvm-cells http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-cells/6a6e0d3/ : SUCCESS in 34m 13s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-full-ceph/cb81abb/ : SUCCESS in 44m 09s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/90/234090/2/check/gate-tempest-dsvm-multinode-full/e2b32f9/ : SUCCESS in 50m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74b18b3753860c1b10e93d3287fa276b9bbe9428","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-15 05:15:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/90/234090/2/check/dsvm-tempest-xen/dd864a8 : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7aa7e27e8d2c015aa067f86ac6154ee011b57a69","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-10-15 05:46:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/90/234090/2/check/dsvm-tempest-cinder/a9ebb5e/ : SUCCESS in 38m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"003c5a7626e1a783aabaf1febe4c5db9fe52f05a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-15 06:15:19.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/2/ : FAILURE in 2h 13m 54s","accounts_in_message":[],"_revision_number":2},{"id":"39ff3c5857ee94bdf95f18657e1da6a1c6190bcd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-15 16:13:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/234090/2 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/234090/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"33c23c8a7b648b51a1cb1ebefe6bb8b970a2a227","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-10-16 05:31:55.000000000","message":"Patch Set 2: Code-Review-1\n\nThe action field of instance_action is exposed by our API, if we changed it, that means something backward incompatible change for our api.","accounts_in_message":[],"_revision_number":2},{"id":"a7de660ff3c0aa349e2c02f2a9f5722421f3eab3","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-16 05:49:32.000000000","message":"Patch Set 2:\n\nNo, Alex. I just changed the string \u0027live-migration\u0027 to \u0027liveMigration\u0027. If you used instance_actions.LIVE_MIGRATION in your code, it is OK. If you are using \u0027live-migration\u0027, I think you should change it.\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"31d38554b3d9dde1baee9ef59802095b3ca95e11","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-10-16 07:30:09.000000000","message":"Patch Set 2:\n\nTang, actually I mean the user of REST API, they won\u0027t reference to nova code.","accounts_in_message":[],"_revision_number":2},{"id":"cbbe73874daba9b1169b1a55629a42060dd15132","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-17 05:50:26.000000000","message":"Patch Set 2:\n\nHi Alex, I\u0027m sorry I don\u0027t quite understand. Do you mean users of REST APIs will use string \u0027live-migration\u0027 directly ?  If so, why do they do that ?","accounts_in_message":[],"_revision_number":2},{"id":"767f3ed20a50e0d9bfae7c1a1594516620efda5e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-10-19 07:20:01.000000000","message":"Patch Set 2:\n\nTangChen, yea, we expose \u0027action\u0027 in the rest API https://github.com/openstack/nova/blob/master/doc/api_samples/os-instance-actions/instance-action-get-resp.json#L3\n\nI think that is kind of api contract, we can\u0027t say user won\u0027t depend on the value of action field, except we said that is just something human readable thing.","accounts_in_message":[],"_revision_number":2},{"id":"934caefaade2e9af7b55e146ac1910666a77fb18","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-19 07:50:13.000000000","message":"Patch Set 2:\n\nHi Alex, thank you for your explanation. Indeed, if we have say this in the doc, we cannot change it easily. So I won\u0027t change the string. I will just change the direct use of the string.\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"291dfb3191df8ce440415d4733224287595c0494","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-19 08:27:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f162125a9c135da88c3c2d4e33f835b2a22b310","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-19 08:45:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/234090/3/check/gate-nova-pep8/1a7d2b3/ : SUCCESS in 7m 16s\n- gate-nova-docs http://docs-draft.openstack.org/90/234090/3/check/gate-nova-docs/0d94aec//doc/build/html/ : SUCCESS in 8m 24s\n- gate-nova-python27 http://logs.openstack.org/90/234090/3/check/gate-nova-python27/a840d78/ : SUCCESS in 9m 33s\n- gate-nova-python34 http://logs.openstack.org/90/234090/3/check/gate-nova-python34/75b9460/ : SUCCESS in 9m 21s\n- gate-nova-tox-functional http://logs.openstack.org/90/234090/3/check/gate-nova-tox-functional/be331e7/ : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":3},{"id":"309690ce5809162dd73e563bac11f4ee14766478","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-10-19 08:47:06.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nnp, a typo in the commit message.","accounts_in_message":[],"_revision_number":3},{"id":"d15e0d21e927b92bff27ffccbe3519058222b5f8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-19 08:47:25.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/234090/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"d48a82264db9ef5032a4de8ec2c143de92c3a10d","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-19 08:56:12.000000000","message":"Patch Set 4: Commit message was updated","accounts_in_message":[],"_revision_number":4},{"id":"77bc650df6907285c45d23442d6fc95cbee3da34","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-19 08:56:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42dd8a8e271d12946cc4226dbd944152fddcfd21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-19 09:12:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/234090/4/check/gate-nova-pep8/fab264c/ : SUCCESS in 10m 08s\n- gate-nova-docs http://docs-draft.openstack.org/90/234090/4/check/gate-nova-docs/67f2d04//doc/build/html/ : SUCCESS in 10m 44s\n- gate-nova-python27 http://logs.openstack.org/90/234090/4/check/gate-nova-python27/2755f96/ : SUCCESS in 14m 00s\n- gate-nova-python34 http://logs.openstack.org/90/234090/4/check/gate-nova-python34/1df9500/ : SUCCESS in 9m 29s\n- gate-nova-tox-functional http://logs.openstack.org/90/234090/4/check/gate-nova-tox-functional/2b21cec/ : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":4},{"id":"8de399f16061189cebe77f7c2fde9774d10667e9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-19 09:14:32.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/234090/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6b54fbdb23d5c27bb43d90db34b61ebba2a5763d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-19 10:49:24.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/4/ : FAILURE in 1h 52m 22s","accounts_in_message":[],"_revision_number":4},{"id":"4cb6427a2d308a28879d68fb18aee78059f3f40a","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-10-19 13:17:00.000000000","message":"Patch Set 5: Commit message was updated","accounts_in_message":[],"_revision_number":5},{"id":"73b3f488c9e9b38302c4a680e955d49215777b3c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2015-10-19 13:17:27.000000000","message":"Patch Set 5: Code-Review+1\n\nJust remove the extra \u0027.\u0027 from commit message title.","accounts_in_message":[],"_revision_number":5},{"id":"509e57978688c1f28afee6c6a32f9f6611557989","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-19 13:36:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/234090/5/check/gate-nova-pep8/fd51d75/ : SUCCESS in 8m 44s\n- gate-nova-docs http://docs-draft.openstack.org/90/234090/5/check/gate-nova-docs/e3b1f92//doc/build/html/ : SUCCESS in 9m 01s\n- gate-nova-python27 http://logs.openstack.org/90/234090/5/check/gate-nova-python27/bd09c87/ : SUCCESS in 15m 38s\n- gate-nova-python34 http://logs.openstack.org/90/234090/5/check/gate-nova-python34/35162fe/ : SUCCESS in 13m 38s\n- gate-nova-tox-functional http://logs.openstack.org/90/234090/5/check/gate-nova-tox-functional/2274ffd/ : SUCCESS in 16m 36s","accounts_in_message":[],"_revision_number":5},{"id":"6800dcb3b4e5c54538ea7ecffd6548e3029657a3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-10-19 13:36:53.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/234090/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"94815719f204f6969ab20875f44ea7fb49d60116","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-19 15:13:59.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/5/ : FAILURE in 1h 56m 23s","accounts_in_message":[],"_revision_number":5},{"id":"e1fedc41621671ce230a8dd27776ff58b3726d8d","author":{"_account_id":14358,"name":"Marian Horban","email":"m.horban@gmail.com","username":"mhorban"},"date":"2015-10-19 15:52:48.000000000","message":"Patch Set 5: Code-Review-1\n\nIt looks to me that it is better to have current version. Because pure string in test checks that value is correct and unchanged.","accounts_in_message":[],"_revision_number":5},{"id":"5832fded6033514d070933d4e6c099731b634803","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-19 16:18:04.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/90/234090/5/14005/index.html : SUCCESS in 55 m 07 s","accounts_in_message":[],"_revision_number":5},{"id":"d5a63ed81f6b30fae018e0b3b95b7f17aeb6ab9b","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-10-20 01:42:22.000000000","message":"Patch Set 5:\n\nHi Marian, \n\nThanks for the advice. That makes sense. Actually, in my first version, I changed the string to \u0027liveMigration\u0027, and the CI did test out the failure. \n\nSo, the principle is:\n- In source code, we should use constants.\n- In test code, we should use strings.\n\nBut a lot of constants are used in test code. Is it necessary to change them all ?","accounts_in_message":[],"_revision_number":5},{"id":"ebb574cf5009d4aea75a74f3f6b6cb52415cb2cc","author":{"_account_id":15882,"name":"Victor Laza","email":"vlaza@cloudbasesolutions.com","username":"vlaza"},"date":"2015-10-20 09:10:21.000000000","message":"Patch Set 3:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":3},{"id":"d6a813782dad49ed09aae0f41250e19877dacaf4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-20 11:38:21.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/3/ : FAILURE in 1h 46m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b5ad24adc0228ad7304d0353b8a79f78b91a0eee","author":{"_account_id":17463,"name":"Nichita Herciu","email":"nherciu@cloudbasesolutions.com","username":"nherciu"},"date":"2015-10-21 11:43:20.000000000","message":"Patch Set 5:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":5},{"id":"428015ebc5b0baa12cf7989d21876813d0e72520","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-21 14:04:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/5/ : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":5},{"id":"4a1f3b218e7dedf2c9ed710dfaa437e67dbfae68","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-21 14:06:07.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7786b6dedeb743a79f35043dbe883559/index.html : SUCCESS in 14m 22s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5e7141a528e5428e9e2f325ca6cfc3ae/index.html : SUCCESS in 14m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4e446305238a4748a45a851a2111cf2c/index.html : SUCCESS in 11m 55s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_percona_user_001:th-percona/e0bf104ffd5e44ada0e2f37116e9a7e5/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e382cec57c934b6db64e803c147917a8/index.html : SUCCESS in 13m 20s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_percona_user_002:th-percona/781634af11db4ad8bf2de8c9d085bf13/index.html : SUCCESS in 14m 34s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"57feea36afcffb3a10b38f3ec226a75684d1ec73","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-11-07 14:29:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"ff5fab06e84fa95720b801b1d5f8b1cc4a76976b","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-11-08 15:55:28.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"ef15eb6ab4716338901605d3e2027e7bd7f822f5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-08 16:11:53.000000000","message":"Patch Set 5:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f1f9991edb354a4f9e96494c7e280181/index.html : SUCCESS in 15m 52s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5b10bcbc2bc7432a941c7d26569e90aa/index.html : SUCCESS in 15m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/43f34912944c407a8c372769672c069b/index.html : SUCCESS in 12m 07s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_percona_user_001:th-percona/5f364604790648ada0f47373ad613ebf/index.html : SUCCESS in 14m 56s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/53246371a4ca4240956b9d66212bfa74/index.html : SUCCESS in 13m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/90/234090/5/check/real-db-upgrade_nova_percona_user_002:th-percona/ab5d2d85081b41d4b2b0ee8d68e1b2ef/index.html : SUCCESS in 14m 19s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"0359f08301bae1318f0d2a7df09397e25968aafa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-08 16:13:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/90/234090/5/check/gate-nova-pep8/c4bf85a/ : SUCCESS in 6m 40s\n- gate-nova-docs http://docs-draft.openstack.org/90/234090/5/check/gate-nova-docs/e91d3f1//doc/build/html/ : SUCCESS in 11m 54s\n- gate-nova-python27 http://logs.openstack.org/90/234090/5/check/gate-nova-python27/adcf4a2/ : SUCCESS in 10m 30s\n- gate-nova-python34 http://logs.openstack.org/90/234090/5/check/gate-nova-python34/31c6f14/ : SUCCESS in 10m 06s\n- gate-nova-tox-functional http://logs.openstack.org/90/234090/5/check/gate-nova-tox-functional/0284a83/ : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":5},{"id":"f0ad4f1637b6fa2135992fb05b3869c9c70b2d73","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-08 17:15:03.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234090/5/ : FAILURE in 1h 19m 08s","accounts_in_message":[],"_revision_number":5},{"id":"a6b481086b44e741589e436762c74ef415a60018","author":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"date":"2015-11-27 18:03:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1c7290f8283bfb75e6eb4f2d192c1cd7569c1642","revisions":{"3ca269b871550a5535514c8ce7a7705f9d0b4549":{"kind":"REWORK","_number":1,"created":"2015-10-13 09:55:21.000000000","uploader":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"ref":"refs/changes/90/234090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/234090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/234090/1"}}},"commit":{"parents":[{"commit":"e11fce969ae912d79175c342307183ca02de7420","subject":"Merge \"Make archive_deleted_rows_for_table private\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e11fce969ae912d79175c342307183ca02de7420"}]}],"author":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-14 09:53:12.000000000","tz":480},"committer":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-14 09:53:12.000000000","tz":480},"subject":"Trivial fix to unify instance_actions string style.","message":"Trivial fix to unify instance_actions string style.\n\nlive-migration should be like liveMigration.\n\nChange-Id: I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748\nSigned-off-by: Tang Chen \u003ctangchen@cn.fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ca269b871550a5535514c8ce7a7705f9d0b4549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ca269b871550a5535514c8ce7a7705f9d0b4549"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e11fce969ae912d79175c342307183ca02de7420","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8a8928e5a5e7235b9b9d66ed8f92a74db8fb37ca":{"kind":"REWORK","_number":2,"created":"2015-10-15 04:00:53.000000000","uploader":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"ref":"refs/changes/90/234090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/234090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/234090/2"}}},"commit":{"parents":[{"commit":"e55130a8d6eb348b004d0c1ee0b39b94c2e0b4b5","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e55130a8d6eb348b004d0c1ee0b39b94c2e0b4b5"}]}],"author":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-14 09:53:12.000000000","tz":480},"committer":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-16 01:38:36.000000000","tz":480},"subject":"Trivial fix to unify instance_actions string style.","message":"Trivial fix to unify instance_actions string style.\n\nlive-migration should be like liveMigration.\n\nChange-Id: I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748\nSigned-off-by: Tang Chen \u003ctangchen@cn.fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a8928e5a5e7235b9b9d66ed8f92a74db8fb37ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a8928e5a5e7235b9b9d66ed8f92a74db8fb37ca"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e55130a8d6eb348b004d0c1ee0b39b94c2e0b4b5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b21768508519ce37ea8c72f6b8062d0d2c7e5332":{"kind":"REWORK","_number":3,"created":"2015-10-19 08:27:40.000000000","uploader":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"ref":"refs/changes/90/234090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/234090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/234090/3"}}},"commit":{"parents":[{"commit":"64e066d11158cc6e88c1b456d6e711b28e2462ce","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64e066d11158cc6e88c1b456d6e711b28e2462ce"}]}],"author":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-14 09:53:12.000000000","tz":480},"committer":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-20 08:15:30.000000000","tz":480},"subject":"Trivial fix to use constant instead of string for instance action.","message":"Trivial fix to use constant instead of string for instance action.\n\nWhen referring to instance action, we should use constat defined in\ninstance_actions.py, but not pure string.\n\nChange-Id: I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748\nSigned-off-by: Tang Chen \u003ctangchen@cn.fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b21768508519ce37ea8c72f6b8062d0d2c7e5332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b21768508519ce37ea8c72f6b8062d0d2c7e5332"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64e066d11158cc6e88c1b456d6e711b28e2462ce","is_merged_in_target_branch":true,"change_id":"I5abfb40a9b0e4cd3c26be205eabde3b61e057e45","change_number":235443,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e7cd469609ba123ad9a2b40c743db190d90b859c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2015-10-19 08:56:12.000000000","uploader":{"_account_id":14937,"name":"Tang Chen","email":"imtangchen@163.com","username":"tangchen"},"ref":"refs/changes/90/234090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/234090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/234090/4"}}},"commit":{"parents":[{"commit":"64e066d11158cc6e88c1b456d6e711b28e2462ce","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64e066d11158cc6e88c1b456d6e711b28e2462ce"}]}],"author":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-14 09:53:12.000000000","tz":480},"committer":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-19 08:56:12.000000000","tz":0},"subject":"Trivial fix to use constant instead of string for instance action.","message":"Trivial fix to use constant instead of string for instance action.\n\nWhen referring to instance action, we should use constant defined in\ninstance_actions.py, but not pure string.\n\nChange-Id: I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748\nSigned-off-by: Tang Chen \u003ctangchen@cn.fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7cd469609ba123ad9a2b40c743db190d90b859c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7cd469609ba123ad9a2b40c743db190d90b859c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64e066d11158cc6e88c1b456d6e711b28e2462ce","is_merged_in_target_branch":true,"change_id":"I5abfb40a9b0e4cd3c26be205eabde3b61e057e45","change_number":235443,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1c7290f8283bfb75e6eb4f2d192c1cd7569c1642":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2015-10-19 13:17:00.000000000","uploader":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"ref":"refs/changes/90/234090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/90/234090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/90/234090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/90/234090/5"}}},"commit":{"parents":[{"commit":"64e066d11158cc6e88c1b456d6e711b28e2462ce","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64e066d11158cc6e88c1b456d6e711b28e2462ce"}]}],"author":{"name":"Tang Chen","email":"tangchen@cn.fujitsu.com","date":"2015-10-14 09:53:12.000000000","tz":480},"committer":{"name":"Alex Xu","email":"hejie.xu@intel.com","date":"2015-10-19 13:17:00.000000000","tz":0},"subject":"Trivial fix to use constant instead of string for instance action","message":"Trivial fix to use constant instead of string for instance action\n\nWhen referring to instance action, we should use constant defined in\ninstance_actions.py, but not pure string.\n\nChange-Id: I03b2249e7e5b46bf055d908cc7a0b8beb4dbd748\nSigned-off-by: Tang Chen \u003ctangchen@cn.fujitsu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c7290f8283bfb75e6eb4f2d192c1cd7569c1642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c7290f8283bfb75e6eb4f2d192c1cd7569c1642"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64e066d11158cc6e88c1b456d6e711b28e2462ce","is_merged_in_target_branch":true,"change_id":"I5abfb40a9b0e4cd3c26be205eabde3b61e057e45","change_number":235443,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
