)]}'
{"id":"openstack%2Fnova~234218","triplet_id":"openstack%2Fnova~master~I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7","project":"openstack/nova","branch":"master","topic":"make_console_use_object","hashtags":[],"change_id":"I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7","subject":"Use instance object instead of db record","status":"MERGED","created":"2015-10-13 13:38:35.000000000","updated":"2016-07-07 01:37:53.000000000","submitted":"2016-07-07 01:37:51.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34fac5628ddcb41a96735f70bfa6bcc73a97538e","_number":234218,"virtual_id_number":234218,"owner":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":2,"date":"2016-07-07 01:37:51.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},{"value":1,"date":"2016-06-14 08:50:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-14 09:30:36.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":1,"date":"2016-06-14 10:00:56.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":2,"date":"2016-07-05 08:32:09.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2016-07-07 00:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-14 08:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2016-07-07 00:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-12-18 17:15:45.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-12-19 04:28:52.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-12-21 00:08:26.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2016-01-05 16:08:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2016-01-06 02:36:18.000000000","updated_by":{"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},"reviewer":{"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},"state":"REVIEWER"},{"updated":"2016-04-08 03:59:23.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-11 07:39:21.000000000","updated_by":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"reviewer":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"state":"REVIEWER"},{"updated":"2016-04-12 11:43:15.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2016-04-12 15:57:26.000000000","updated_by":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"reviewer":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"state":"REVIEWER"},{"updated":"2016-04-14 12:59:48.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-14 13:43:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-14 14:50:55.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-14 14:55:16.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-14 15:21:54.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-14 15:55:14.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-04-15 19:23:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-06 09:50:31.000000000","updated_by":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"reviewer":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"state":"REVIEWER"},{"updated":"2016-06-14 08:50:27.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-14 09:30:36.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-14 10:00:56.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-07-05 08:32:09.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-07-07 00:01:55.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2016-07-07 01:37:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c0d5bd6d2d12ec2d60311efda4a03d78e87c0a4a","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-10-13 13:38:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6386016c9527e448bdd922c8fc627d6d07923db4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-10-13 14:32:02.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/1/check/check-ibm-tempest-dsvm-full/b4bc86a/ : FAILURE in 52m 38s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/1/check/check-ibm-tempest-dsvm-postgres-full/8fe84d5/ : SUCCESS in 48m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9e6c33c914a1c3fe26864501518bcbeb70058f7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-10-13 14:43:45.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/1/13554/index.html : SUCCESS in 01 h 00 m 02 s","accounts_in_message":[],"_revision_number":1},{"id":"9c05c1710e5163205b4e671d1c9f1346570162c7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-13 15:39:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/1/check/dsvm-tempest-xen/c0815eb : FAILURE in 1h 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ede83d1de4ec9f90e3a800b93cff0cc301b4b64d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-13 17:15:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/1/check/gate-nova-pep8/28fb8d8/ : SUCCESS in 8m 53s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/1/check/gate-nova-docs/40d33d7//doc/build/html/ : SUCCESS in 15m 09s\n- gate-nova-python27 http://logs.openstack.org/18/234218/1/check/gate-nova-python27/a4ac176/ : SUCCESS in 14m 55s\n- gate-nova-python34 http://logs.openstack.org/18/234218/1/check/gate-nova-python34/f26a1a9/ : SUCCESS in 12m 33s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-full/1784531/ : SUCCESS in 48m 40s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-postgres-full/3aab338/ : SUCCESS in 40m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-neutron-full/c17e8ee/ : SUCCESS in 58m 42s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/1/check/gate-grenade-dsvm/74ef49f/ : SUCCESS in 56m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-large-ops/216262f/ : SUCCESS in 28m 02s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-neutron-large-ops/60d8354/ : SUCCESS in 27m 04s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/1/check/gate-nova-tox-functional/fd032ee/ : SUCCESS in 12m 10s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/18/234218/1/check/gate-grenade-dsvm-partial-ncpu/3a68b3d/ : SUCCESS in 44m 09s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/443ed1c/ : SUCCESS in 36m 28s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-nova-v20-api/70f54f8/ : SUCCESS in 33m 59s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-nova-v20-api-legacy/0e3053e/ : SUCCESS in 44m 15s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-cells/9360d17/ : SUCCESS in 37m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-full-ceph/4feb0f2/ : SUCCESS in 57m 59s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/1/check/gate-tempest-dsvm-multinode-full/1e8b942/ : FAILURE in 52m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41608aafea3b672285315478347dac411f0d8ef4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-10-13 17:45:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/18/234218/1/check/dsvm-tempest-cinder/110849d/ : FAILURE in 29m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"e3f3fb60fa2543a4a54ffb7ac8037fae5c11fbc3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-10-13 18:46:07.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b0578a3099324692a967fc9c67823543/index.html : SUCCESS in 13m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/600c0303b20d49028e6b8f827f496e53/index.html : SUCCESS in 15m 21s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5a813a4978f946e79f43489189973180/index.html : SUCCESS in 14m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/1/check/real-db-upgrade_nova_percona_user_001:th-percona/e37ec385bbfd4eed882f02c16f1a3f4f/index.html : SUCCESS in 15m 42s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/277f377ea0a24147b6bc226917548ac0/index.html : SUCCESS in 15m 32s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/1/check/real-db-upgrade_nova_percona_user_002:th-percona/2b6c7db956834723a9801d9a048de45e/index.html : SUCCESS in 12m 18s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"94745bf16f674e099453e4dad5147c20fe73e77e","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-10-13 19:30:32.000000000","message":"Patch Set 1:\n\nxen: recheck","accounts_in_message":[],"_revision_number":1},{"id":"ceac6e66d11e7298889d9659420b7ddc7f2e5843","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-10-13 20:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/1/check/dsvm-tempest-xen/340a555 : SUCCESS in 1h 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"31af48cfa57d5068b4706b9ae025d322b7561dde","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-10-14 00:38:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/1/ : FAILURE in 2h 29m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6da18beb8f2fe5b40f0281d4f5815620495bcb51","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-10-14 06:01:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/234218/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/234218/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cbcb91a070106c420a8ad83193f2b0320279969f","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-11-24 10:51:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"11257974b4e4f515b1504eb51eafd1745d8342e9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-24 11:27:59.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/234218/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"91f1ae44e3f722b11ec780419d3317c84d51ddaa","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-24 11:44:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/2/ : FAILURE in 47m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4b9dd3180478a80016177a283e938037ed028a37","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-24 12:01:20.000000000","message":"Patch Set 2: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/2/16921/index.html : FAILURE in 01 h 05 m 26 s","accounts_in_message":[],"_revision_number":2},{"id":"99f2423b66a1e600df44bb4aaef73870af822b49","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-11-24 12:02:03.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/2/check/check-ibm-tempest-dsvm-full/db4526f/ : SUCCESS in 1h 06m 28s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/2/check/check-ibm-tempest-dsvm-postgres-full/ce8136a/ : SUCCESS in 1h 00m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20a3c3879f6a0afd4f99546dc30eb4e0aaf6016e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-24 12:16:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/2/check/dsvm-tempest-xen/87edbd0 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":2},{"id":"af252e5eda549242ce841efea30ab50fede4af29","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-24 12:24:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/2/check/gate-nova-pep8/459c596/ : SUCCESS in 9m 36s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/2/check/gate-nova-docs/3d0eb7f//doc/build/html/ : SUCCESS in 9m 36s\n- gate-nova-python27 http://logs.openstack.org/18/234218/2/check/gate-nova-python27/74af49f/ : SUCCESS in 11m 13s\n- gate-nova-python34 http://logs.openstack.org/18/234218/2/check/gate-nova-python34/2dd9058/ : FAILURE in 12m 46s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-full/d3426b6/ : SUCCESS in 53m 34s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-postgres-full/77e89b6/ : SUCCESS in 52m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-neutron-full/7e08da7/ : SUCCESS in 1h 16m 09s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/2/check/gate-grenade-dsvm/23044a2/ : SUCCESS in 47m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-large-ops/69ed880/ : SUCCESS in 41m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-neutron-large-ops/a3a2394/ : SUCCESS in 32m 44s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/2/check/gate-nova-tox-functional/c60cffb/ : SUCCESS in 15m 59s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/2/check/gate-grenade-dsvm-multinode/cb4aee0/ : SUCCESS in 47m 57s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/544134e/ : SUCCESS in 51m 13s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-nova-v20-api/3cc8d9c/ : SUCCESS in 46m 22s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-nova-v20-api-legacy/3397ea5/ : SUCCESS in 40m 29s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-cells/3dd2d87/ : SUCCESS in 51m 45s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-full-ceph/6b205da/ : SUCCESS in 41m 18s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/2/check/gate-tempest-dsvm-multinode-full/35b9f16/ : FAILURE in 57m 44s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/18/234218/2/check/gate-compute-api-guide/811ccb6//api-guide/build/html/ : SUCCESS in 9m 24s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/18/234218/2/check/gate-nova-python27-constraints/4403f31/ : SUCCESS in 14m 49s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/18/234218/2/check/gate-nova-python34-constraints/c64c986/ : SUCCESS in 13m 07s (non-voting)\n- gate-nova-docs-constraints http://logs.openstack.org/18/234218/2/check/gate-nova-docs-constraints/68d263e/ : SUCCESS in 15m 03s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/18/234218/2/check/gate-nova-pep8-constraints/94281df/ : SUCCESS in 9m 08s (non-voting)\n- gate-nova-tox-functional-constraints http://logs.openstack.org/18/234218/2/check/gate-nova-tox-functional-constraints/97fb1d8/ : FAILURE in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ae0864317b08294a9250e184f28444c7d25faa7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-24 14:37:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/234218/2/Nova-ML2-Sriov : SUCCESS in 1h 14m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"43f4e964d61f9acd86c09f97bb45c7cfdd33bb25","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2015-11-24 20:57:04.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThe code\u0027s failing unit tests (which I didn\u0027t see before I started reviewing) but the ideas good, IMO. Fix the commit and broken tests and I\u0027ll happily +1.","accounts_in_message":[],"_revision_number":2},{"id":"274e4d78055aa7268f9d4a562cf473739ac3b3e0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-25 09:03:23.000000000","message":"Patch Set 2:\n\nBuild failed. Logs: http://208.91.1.172/logs/234218/2 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-dsvm-tempest http://208.91.1.172/logs/234218/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"33375c60d0ab336b40b1b851666466f789e04a1a","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-11-25 12:18:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf2940460b01b591b554a01873509a47f510b6b3","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-11-25 12:18:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9b132e0558fc055d62d5cf7dc340ce7241dfde0","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-11-25 12:42:56.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/aec4882517a24647a9cf3e46cfdaaf8e/index.html : SUCCESS in 14m 51s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5686655f09a3455c8a29f1ccc0559a63/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8c3de1915d4a4367a36ee2c163aec862/index.html : SUCCESS in 14m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/3/check/real-db-upgrade_nova_percona_user_001:th-percona/135ebfb9301141759048dc7d9713f46e/index.html : SUCCESS in 16m 35s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/687f7fa2df194796b16d8b7cfcc73234/index.html : SUCCESS in 15m 18s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/3/check/real-db-upgrade_nova_percona_user_002:th-percona/5f0905ac2862474ba5f004962afa644f/index.html : SUCCESS in 15m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"835c0ada49f2abc934d2619085494ac55f80127f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-11-25 13:20:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/3/ : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":3},{"id":"eff670bcac9b8be7de5b0273d75eebcc34d80322","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-11-25 13:22:22.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/3/check/check-ibm-tempest-dsvm-full/2c6a31b/ : SUCCESS in 1h 02m 35s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/3/check/check-ibm-tempest-dsvm-postgres-full/cbbf501/ : SUCCESS in 1h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd42196f9718c1c39d3d45d08c67df3cdc81673d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-11-25 13:26:11.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/234218/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"19f6fe20e0b6b7c03e962146887f2a6bd4f509f4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-11-25 13:28:46.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/3/17065/index.html : SUCCESS in 54 m 09 s","accounts_in_message":[],"_revision_number":3},{"id":"22bfcd42b413fd930c3620ad6a6cdbabf4cae664","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-11-25 13:36:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/3/check/dsvm-tempest-xen/ca6025c : SUCCESS in 1h 12m 21s","accounts_in_message":[],"_revision_number":3},{"id":"330132a1528c9270044ca9c3c5fa6b1e65c7a73d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-25 13:43:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/3/check/gate-nova-pep8/b901c20/ : SUCCESS in 9m 05s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/3/check/gate-nova-docs/21c0895//doc/build/html/ : SUCCESS in 15m 22s\n- gate-nova-python27 http://logs.openstack.org/18/234218/3/check/gate-nova-python27/ed123a4/ : SUCCESS in 15m 53s\n- gate-nova-python34 http://logs.openstack.org/18/234218/3/check/gate-nova-python34/da7f29d/ : SUCCESS in 12m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-full/9e32c62/ : SUCCESS in 44m 11s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-postgres-full/d1d0fd1/ : SUCCESS in 44m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-neutron-full/20b2594/ : SUCCESS in 56m 19s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/3/check/gate-grenade-dsvm/6603aad/ : SUCCESS in 39m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-large-ops/5e2502e/ : SUCCESS in 29m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-neutron-large-ops/e299f76/ : SUCCESS in 26m 46s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/3/check/gate-nova-tox-functional/a7e97ba/ : SUCCESS in 12m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/3/check/gate-grenade-dsvm-multinode/244c11c/ : SUCCESS in 1h 05m 26s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/7f0ec71/ : SUCCESS in 40m 43s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-nova-v20-api/1d82d25/ : SUCCESS in 50m 30s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-nova-v20-api-legacy/cab8fde/ : SUCCESS in 45m 51s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-cells/6631bf8/ : SUCCESS in 45m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-full-ceph/32ef4e1/ : SUCCESS in 48m 11s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/3/check/gate-tempest-dsvm-multinode-full/d19c3e9/ : FAILURE in 1h 14m 51s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/18/234218/3/check/gate-compute-api-guide/af0d9c8//api-guide/build/html/ : SUCCESS in 8m 35s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/18/234218/3/check/gate-nova-python27-constraints/a8cd47f/ : SUCCESS in 14m 07s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/18/234218/3/check/gate-nova-python34-constraints/720ca1c/ : SUCCESS in 13m 06s (non-voting)\n- gate-nova-docs-constraints http://logs.openstack.org/18/234218/3/check/gate-nova-docs-constraints/9e1ead7/ : SUCCESS in 14m 44s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/18/234218/3/check/gate-nova-pep8-constraints/ef7c7a1/ : SUCCESS in 10m 01s (non-voting)\n- gate-nova-tox-functional-constraints http://logs.openstack.org/18/234218/3/check/gate-nova-tox-functional-constraints/1a7813d/ : FAILURE in 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5dbbb84e04c60a563300e57201dd9f8a76a3b844","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-11-25 14:52:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/18/234218/3/check/check-dsvm-vz7-tempest-exe-minimal/5fd76f7/ : FAILURE in 53m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"4ca5fbb6194cce8daba725864aa7f100d5f5f5e5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-11-25 14:57:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/18/234218/3/check/dsvm-tempest-cinder/5e62cd7/ : SUCCESS in 47m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"fe24dda95b788937b5c3d2c9a910dd1c0d7bda47","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-11-25 16:48:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/234218/3/Nova-ML2-Sriov : SUCCESS in 57m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"d989cfb3d456b6ee88ebc3174c73263998876a89","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-11-26 17:39:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/234218/3 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/234218/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"4b4493f5c7b3c11d8ed6beb5933f2c5fcd1ea303","author":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"date":"2015-11-27 09:47:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"068639e43b7327254e37f36f3c449c682a5bdbab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-12 00:19:41.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"f02a852191e827b90694fa461ec7603c274f515b","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-12-15 13:34:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0682fc56ec90e3198cfb1cfa8e4a67d0b4fa9833","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-15 14:31:41.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/4/check/check-ibm-tempest-dsvm-full/c2fae9c/ : SUCCESS in 50m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a780c01aa4e4c9b077af99d554e1058f827b6661","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-15 15:20:32.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/308ade1439af45c2877023efe60a1619/index.html : SUCCESS in 12m 52s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e0879a6240c94968945e1bb670753940/index.html : SUCCESS in 15m 45s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/41f08be5a36a4fe084f57c7df11fc60b/index.html : SUCCESS in 12m 43s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/4/check/real-db-upgrade_nova_percona_user_001:th-percona/019609c058bd40518b9124e0cedead0a/index.html : SUCCESS in 16m 23s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/0c63f966ffd94493a8c226a4a943ae4f/index.html : SUCCESS in 14m 44s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/4/check/real-db-upgrade_nova_percona_user_002:th-percona/f80bbed2d67e4268ab37a56a22b1a82c/index.html : SUCCESS in 13m 40s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"345daef1f2a2bdcf791dc246f218815df17dfbe2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-15 15:29:23.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/4/19726/index.html : SUCCESS in 58 m 23 s","accounts_in_message":[],"_revision_number":4},{"id":"8d8d08dfc3356eb15b5f99fd5d598fb33d5237a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 15:31:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/4/check/gate-nova-pep8/4cd032c/ : SUCCESS in 10m 03s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/4/check/gate-nova-docs/06d1db6//doc/build/html/ : SUCCESS in 14m 47s\n- gate-nova-python27 http://logs.openstack.org/18/234218/4/check/gate-nova-python27/f90f51d/ : SUCCESS in 14m 10s\n- gate-nova-python34 http://logs.openstack.org/18/234218/4/check/gate-nova-python34/11132ff/ : SUCCESS in 14m 46s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-full/99caf30/ : SUCCESS in 1h 08m 11s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-postgres-full/c33574b/ : SUCCESS in 53m 55s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-neutron-full/3359af7/ : SUCCESS in 1h 17m 11s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/4/check/gate-grenade-dsvm/796a720/ : SUCCESS in 49m 53s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-large-ops/fdf469e/ : SUCCESS in 37m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-neutron-large-ops/b6b65e1/ : SUCCESS in 31m 39s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/4/check/gate-nova-tox-functional/85ccaa0/ : SUCCESS in 16m 56s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/4/check/gate-grenade-dsvm-multinode/b449d1c/ : SUCCESS in 1h 00m 42s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f186b48/ : FAILURE in 46m 26s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-nova-v20-api/12bffbf/ : SUCCESS in 52m 44s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-nova-v20-api-legacy/eb6cdd2/ : SUCCESS in 34m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-cells/a0ac656/ : SUCCESS in 39m 19s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-full-ceph/dc6b1dd/ : SUCCESS in 46m 41s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/bed7528/ : SUCCESS in 46m 00s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/4/check/gate-tempest-dsvm-multinode-full/24ff000/ : FAILURE in 1h 02m 51s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/18/234218/4/check/gate-compute-api-guide/2561214//api-guide/build/html/ : SUCCESS in 8m 46s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/18/234218/4/check/gate-nova-python27-constraints/66e281f/ : SUCCESS in 14m 53s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/18/234218/4/check/gate-nova-python34-constraints/d7c31db/ : SUCCESS in 13m 50s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/18/234218/4/check/gate-nova-docs-constraints/16867ce//doc/build/html/ : SUCCESS in 15m 03s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/18/234218/4/check/gate-nova-pep8-constraints/2dcfbb1/ : SUCCESS in 9m 23s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/18/234218/4/check/gate-nova-tox-constraints-functional-constraints/70940c5/ : SUCCESS in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0ade1777a0ab8992dd9f4f4079262167a225d8cc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-15 15:34:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/4/check/dsvm-tempest-xen/65e67db : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5f358b205a60f4e471e5c187314c47d24d6105b6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 16:27:43.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/4/ : FAILURE in 1h 54m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b5d96202c36fce6c2dbfacee2b5e0b6a84b5f1a3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-15 19:08:33.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/234218/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"605f45d69af633791cbe556bfd941fa0c89ed8ac","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 22:02:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/18/234218/4/check/check-dsvm-vz7-tempest-exe-minimal/a4d89bb/ : SUCCESS in 50m 14s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"6d53c87586b89ba83dd12d12cd9606229ca03082","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-15 22:35:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/18/234218/4/check/dsvm-tempest-cinder/414d1ed/ : SUCCESS in 56m 32s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"abfa3e007eb5488a2e5ca70db8331d72f4b53977","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-16 03:54:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/234218/4/Nova-ML2-Sriov : SUCCESS in 57m 23s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"1efe466992d3d7d9ab804c51e12b905f83fe000a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2015-12-16 04:26:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nThe fix is good. But the commit message should be corrected.","accounts_in_message":[],"_revision_number":4},{"id":"f198cd95a9d60d836d4d14eb9cddc218bbb1f9d8","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-12-18 15:22:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"591e00875b6e790d27dd4834098c9a592421b927","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-18 16:11:00.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/5/check/check-ibm-tempest-dsvm-full/dec3e60/ : SUCCESS in 47m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cf541b24eee3a268e8775ea6ff8528a8204b93f4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-18 16:33:07.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/5/20431/index.html : SUCCESS in 01 h 03 m 46 s","accounts_in_message":[],"_revision_number":5},{"id":"db49f1558fd1cc3e3e233c58086f215135590333","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-18 16:48:52.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b92e7b41da8f413eaa4041f69d1f0aaf/index.html : SUCCESS in 14m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/dd16d14fc98141f5bba2f48150de16db/index.html : SUCCESS in 14m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/bf7bb2e824734953a5a9b29b46ef411a/index.html : SUCCESS in 14m 31s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/5/check/real-db-upgrade_nova_percona_user_001:th-percona/0232f70fb51944aa8c89a9265d9d9688/index.html : SUCCESS in 16m 25s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5e6ba85262154140aa4e3669c123fea9/index.html : SUCCESS in 13m 35s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/5/check/real-db-upgrade_nova_percona_user_002:th-percona/51999aa8029d45418ef24d3618cecd23/index.html : SUCCESS in 15m 33s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"8f277121ce9566ce6538bd26a0e6a42b1db14877","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-18 16:59:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/5/ : SUCCESS in 1h 36m 09s","accounts_in_message":[],"_revision_number":5},{"id":"7436aa415539246e7ff7601602f9a8ef113a37de","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-18 17:05:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/5/check/dsvm-tempest-xen/e38ce9f : SUCCESS in 1h 12m 36s","accounts_in_message":[],"_revision_number":5},{"id":"ef86c08a98aa0e129946d54c307380e68557bd03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-18 17:14:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/5/check/gate-nova-pep8/8cfb782/ : SUCCESS in 8m 52s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/5/check/gate-nova-docs/13f9ac7//doc/build/html/ : SUCCESS in 13m 28s\n- gate-nova-python27 http://logs.openstack.org/18/234218/5/check/gate-nova-python27/32e4697/ : SUCCESS in 13m 22s\n- gate-nova-python34 http://logs.openstack.org/18/234218/5/check/gate-nova-python34/182d50c/ : SUCCESS in 12m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-full/3d45dd0/ : SUCCESS in 40m 55s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-postgres-full/8b4618a/ : SUCCESS in 39m 40s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-neutron-full/235b08e/ : FAILURE in 1h 07m 49s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/5/check/gate-grenade-dsvm/b7adef7/ : SUCCESS in 53m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-large-ops/9e28cf9/ : SUCCESS in 36m 15s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-neutron-large-ops/ba62b4a/ : SUCCESS in 31m 28s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/5/check/gate-nova-tox-functional/59de2fe/ : SUCCESS in 14m 57s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/5/check/gate-grenade-dsvm-multinode/2feb81a/ : SUCCESS in 48m 39s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8858703/ : SUCCESS in 45m 58s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-nova-v20-api/2c5ce60/ : SUCCESS in 47m 47s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-nova-v20-api-legacy/d2850af/ : SUCCESS in 48m 50s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-cells/17a4d61/ : SUCCESS in 39m 28s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-full-ceph/1b838d6/ : SUCCESS in 56m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/d086556/ : SUCCESS in 42m 50s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-multinode-full/efa27f0/ : FAILURE in 1h 03m 01s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/18/234218/5/check/gate-compute-api-guide/eb51045//api-guide/build/html/ : SUCCESS in 7m 58s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/18/234218/5/check/gate-nova-commit-message/d4e5887/ : FAILURE in 53s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-python27-constraints/fcb836b/ : SUCCESS in 14m 58s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-python34-constraints/afcef95/ : SUCCESS in 12m 27s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/18/234218/5/check/gate-nova-docs-constraints/bc632a0//doc/build/html/ : SUCCESS in 15m 22s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-pep8-constraints/38a7db7/ : SUCCESS in 8m 58s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-tox-constraints-functional-constraints/b8d8976/ : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9ff06b61220afbb97a952bdc5a4f5538d4d31f6b","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-12-18 17:15:45.000000000","message":"Patch Set 5:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1509004\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":5},{"id":"3264c7995bb8d22252cf87fac6399e3154756c48","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2015-12-18 17:19:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2884057fd4232fe81234d19c5c41b93a010bbdb0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-18 17:38:11.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/234218/5/check/check-dsvm-tempest-vz7-exe-minimal/000bcbd : FAILURE in 32m 41s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"591e2cd57862a33d105e7db96ca5e992fcc72e80","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-18 18:16:48.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/5/check/check-ibm-tempest-dsvm-full/0018c7a/ : SUCCESS in 48m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"054cdf7050063f91c4f1a19576a31eab188e345c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-18 19:55:43.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/5/ : FAILURE in 1h 55m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d1e5a315e8ed86b97c74554653941472f71e2e0a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-18 20:50:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/5/check/gate-nova-pep8/91735dc/ : SUCCESS in 7m 23s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/5/check/gate-nova-docs/7accc7c//doc/build/html/ : SUCCESS in 12m 01s\n- gate-nova-python27 http://logs.openstack.org/18/234218/5/check/gate-nova-python27/1ec0f1d/ : SUCCESS in 11m 20s\n- gate-nova-python34 http://logs.openstack.org/18/234218/5/check/gate-nova-python34/1ffd779/ : SUCCESS in 14m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-full/00d71b5/ : SUCCESS in 50m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-postgres-full/0cafd51/ : SUCCESS in 46m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-neutron-full/90e5676/ : SUCCESS in 1h 19m 37s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/5/check/gate-grenade-dsvm/97f4123/ : SUCCESS in 46m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-large-ops/cf3b898/ : SUCCESS in 33m 16s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-neutron-large-ops/1e1094a/ : SUCCESS in 30m 28s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/5/check/gate-nova-tox-functional/dd2143e/ : SUCCESS in 12m 58s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/5/check/gate-grenade-dsvm-multinode/4bf7642/ : SUCCESS in 54m 02s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8162abb/ : FAILURE in 47m 52s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-nova-v20-api/0a9f59b/ : SUCCESS in 47m 46s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-nova-v20-api-legacy/419b8f5/ : SUCCESS in 46m 49s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-cells/074aec1/ : SUCCESS in 39m 56s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-full-ceph/26c5994/ : SUCCESS in 57m 00s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/0261ed8/ : SUCCESS in 1h 06m 17s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/5/check/gate-tempest-dsvm-multinode-full/32755c5/ : SUCCESS in 45m 20s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/18/234218/5/check/gate-compute-api-guide/964eafb//api-guide/build/html/ : SUCCESS in 6m 56s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/18/234218/5/check/gate-nova-commit-message/5fd7125/ : FAILURE in 1m 11s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-python27-constraints/9e0aa76/ : SUCCESS in 12m 44s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-python34-constraints/ed16fa5/ : SUCCESS in 10m 54s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/18/234218/5/check/gate-nova-docs-constraints/72aea3f//doc/build/html/ : SUCCESS in 12m 41s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-pep8-constraints/5b5f756/ : SUCCESS in 7m 46s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/18/234218/5/check/gate-nova-tox-constraints-functional-constraints/b7e30f5/ : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f1f30b6f6fcbaf49e42b97e1a8bd8482145dcfa","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-18 23:22:59.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/234218/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"fb2873f241232150dbadd9c904494657d216d213","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-18 23:33:47.000000000","message":"Patch Set 5:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/234218/5/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"0b3063e84167d0b6436bbf03066b8ec1f0e0b5d3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2015-12-19 04:28:52.000000000","message":"Patch Set 5:\n\nBuild Successful.\n\n* compute-nfv-flavors http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/234218/5/compute-nfv-flavors : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"e2fd284b8d17b904693aa34bea1610fbad0914cc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-19 06:04:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/18/234218/5/check/dsvm-tempest-kvm/f367c4b : SUCCESS in 1h 03m 35s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"6c2fd1d640a2a214a148316f96703d1a7f5d7cd9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-19 13:24:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/234218/5/Nova-ML2-Sriov : SUCCESS in 1h 00m 06s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"114736d39d8c82a2bfcea09f2bda49f158bb8b4a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2015-12-21 00:08:26.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6174daaecbd4014c0b816e895a6996c1dc80d753","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-22 13:15:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/234218/5 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/385/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c14a976e0a82a7838419ef5aebe5e72a5db57eac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-01-05 16:08:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cb35aa9457fdc752eda8585a43ec7003c7a6c6cc","author":{"_account_id":14762,"name":"Zhihai Song","email":"zhhsng@gmail.com","username":"ZhihaiSong"},"date":"2016-01-06 02:36:18.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"29d5f00e8580053703ec2a1328905f9ca16ad9f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 07:34:26.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"a6ccc3e96bc165ea6b4af16f0b6e92eea461d3b1","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-03-07 12:10:30.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c08195505836458cdfd22ad355ca43108c6c62c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-07 13:16:10.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/6/check/check-ibm-tempest-dsvm-full/7306b03/ : SUCCESS in 1h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"954591f48a30832966ef19aa4b3ad4ab4863633b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 13:32:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/234218/6/check/gate-nova-pep8/166ca49/ : SUCCESS in 2m 34s\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/6/check/gate-nova-docs/ab42551//doc/build/html/ : SUCCESS in 9m 23s\n- gate-nova-python27 http://logs.openstack.org/18/234218/6/check/gate-nova-python27/02f5093/ : SUCCESS in 7m 54s\n- gate-nova-python34 http://logs.openstack.org/18/234218/6/check/gate-nova-python34/faa8593/ : SUCCESS in 29m 46s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-full/73ed515/ : SUCCESS in 48m 30s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-postgres-full/1af3582/ : SUCCESS in 39m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-neutron-full/f975bae/ : SUCCESS in 1h 15m 10s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/6/check/gate-grenade-dsvm/a6676be/ : SUCCESS in 38m 04s\n- gate-nova-tox-functional http://logs.openstack.org/18/234218/6/check/gate-nova-tox-functional/fc96db3/ : SUCCESS in 9m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/6/check/gate-grenade-dsvm-multinode/4d906af/ : SUCCESS in 47m 27s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b8fe2de/ : SUCCESS in 42m 34s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-cells/a0bd4de/ : SUCCESS in 49m 50s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-full-ceph/c665f94/ : SUCCESS in 40m 00s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/aded8b7/ : SUCCESS in 52m 14s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/6/check/gate-tempest-dsvm-multinode-full/ba9c5c7/ : FAILURE in 1h 09m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bb129d9bf86537ea39c627cea6b1ae363e0710f7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 13:43:43.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/6/ : FAILURE in 15m 04s","accounts_in_message":[],"_revision_number":6},{"id":"befe385a2eb452776687634bbc56ec6bc169973a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-07 13:44:40.000000000","message":"Patch Set 6: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/6/8474/index.html : SUCCESS in 01 h 13 m 20 s","accounts_in_message":[],"_revision_number":6},{"id":"7a3918801cf6d8b11d2db00691e9e4f03e4bb3dc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-07 13:51:37.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f1c7fe3bf6734cb9ba0f95ccb0db3fa9/index.html : SUCCESS in 12m 52s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ed1449e4e89646be973fa578eca65c0b/index.html : SUCCESS in 16m 42s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/76dde87ddca0440fa7838c6af185262e/index.html : SUCCESS in 15m 50s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/6/check/real-db-upgrade_nova_percona_user_001:th-percona/c020626f5f9c4d70bee4f2c68d47a968/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/344a32f7796e43c993616297430af5c1/index.html : SUCCESS in 20m 16s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/6/check/real-db-upgrade_nova_percona_user_002:th-percona/6c0736dbd168493c93ed07373d790def/index.html : SUCCESS in 12m 55s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"bca487a2a825569d9ce2af630bbf7686c4a85632","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-07 14:09:37.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/234218/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"fd56840e8e0eef2f9982a54ea47b822c6d2b2195","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-07 14:58:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/6/check/dsvm-tempest-xen/dcbf57f : FAILURE in 1h 21m 01s","accounts_in_message":[],"_revision_number":6},{"id":"604d1c9a747dd0fbcecdf4e4e5c097ab4fdc99df","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 21:57:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/234218/6/check/check-dsvm-tempest-vz7-exe-minimal/492e9be : SUCCESS in 1h 20m 01s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"d34446d40fdc751ba0b1431b794d5c6c04176362","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-03-09 10:52:53.000000000","message":"Patch Set 6:\n\nxen: recheck","accounts_in_message":[],"_revision_number":6},{"id":"58ac1b3fdf3bb3ce6bba61be9d145355fcab8a23","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-09 12:37:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/6/check/dsvm-tempest-xen/9b84baf : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":6},{"id":"98383133dceb8718bdcd47f60e064e226e53fc26","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-12 04:21:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/234218/6/2846 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"f06fe566ad345c36d42a0afffa7fa33dc85137f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 11:14:22.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"a2fa647d620b200d9b65d4f6882af7336e5bcf64","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-04-07 19:01:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"843b8765cc68ffd230c6c5171aafa22abb757c3f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-07 19:07:40.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/7/check/check-ibm-tempest-dsvm-full/ffe7a18/ : FAILURE in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9d6d93e6412357697577de87b0302a62a0411a39","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-07 19:21:25.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/434be07b2b3b42d2a400384a6b9de5a4/index.html : SUCCESS in 11m 53s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e91b7fdd8f2a48faba6d5c4626b1427d/index.html : SUCCESS in 16m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ba764a25083c4216bd44f67b95ff697a/index.html : SUCCESS in 12m 35s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/7/check/real-db-upgrade_nova_percona_user_001:th-percona/bf809defebab41dcba4e24afc6a3d811/index.html : SUCCESS in 14m 17s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"30baf9ab370f80ca19db4aee6c72c9ecb7c75cef","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-07 19:29:23.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/234218/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"7ae863e5d23f17482b457a79b49a9c9923982ac8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-07 20:26:08.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/7/11587/index.html : SUCCESS in 01 h 22 m 18 s","accounts_in_message":[],"_revision_number":7},{"id":"edc442b54008f3b1a8d5c67e634222e012760bce","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-07 20:40:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/7/ : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":7},{"id":"6c52024b685bab6f3cd0c274ade1e028edfd1b1c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 21:29:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/7/check/gate-nova-docs/8ec40f8//doc/build/html/ : SUCCESS in 2m 55s\n- gate-nova-pep8 http://logs.openstack.org/18/234218/7/check/gate-nova-pep8/1cded96/ : SUCCESS in 2m 56s\n- gate-nova-python27-db http://logs.openstack.org/18/234218/7/check/gate-nova-python27-db/c3425a8/ : FAILURE in 9m 23s\n- gate-nova-python34-db http://logs.openstack.org/18/234218/7/check/gate-nova-python34-db/f06e8bb/ : FAILURE in 14m 21s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-full/cdfaf36/ : SUCCESS in 44m 55s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-postgres-full/d20bdd8/ : FAILURE in 1h 10m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-neutron-full/8c1b520/ : SUCCESS in 1h 01m 09s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/7/check/gate-grenade-dsvm/816b917/ : SUCCESS in 24m 22s\n- gate-nova-tox-db-functional http://logs.openstack.org/18/234218/7/check/gate-nova-tox-db-functional/13ebbb2/ : FAILURE in 5m 47s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/7/check/gate-grenade-dsvm-multinode/e2f0877/ : SUCCESS in 28m 02s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/cca3bc1/ : SUCCESS in 32m 32s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-cells/b52cc37/ : SUCCESS in 25m 42s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/f6acc36/ : SUCCESS in 41m 49s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-multinode-full/d89e7f7/ : FAILURE in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"260d657f8123461ef8718d386cb4b83aadde54c9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-07 21:35:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/7/check/dsvm-tempest-xen/3060117 : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":7},{"id":"9ccbba651566e65b923d2023998f098090b50d2d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-04-08 03:59:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/18/234218/7/check/dsvm-tempest-kvm/a9a75dc : FAILURE in 15m 30s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"c1caaa0cd14f3108a53c834ba966f57adbc647b4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-08 10:48:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/234218/7/check/check-dsvm-tempest-vz7-exe-minimal/e326bc0 : SUCCESS in 1h 29m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"10d5cd3545786fdd8d5300c1b2c4766c07d592ae","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-04-08 10:58:16.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0447564743a67d7f0e4a150951abaaf7b4a4435a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-08 12:04:11.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/7/check/check-ibm-tempest-dsvm-full/3f9226f/ : SUCCESS in 1h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ffeeb03c8eff519fbf17576922771edbaa505ba4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-08 12:09:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/7/check/gate-nova-docs/7f5eb76//doc/build/html/ : SUCCESS in 2m 36s\n- gate-nova-pep8 http://logs.openstack.org/18/234218/7/check/gate-nova-pep8/b20ddc7/ : SUCCESS in 4m 00s\n- gate-nova-python27-db http://logs.openstack.org/18/234218/7/check/gate-nova-python27-db/57dab94/ : SUCCESS in 8m 41s\n- gate-nova-python34-db http://logs.openstack.org/18/234218/7/check/gate-nova-python34-db/25f1bc5/ : SUCCESS in 15m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-full/d2ad71d/ : SUCCESS in 1h 01m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-postgres-full/6a9bef4/ : SUCCESS in 33m 55s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-neutron-full/330f101/ : SUCCESS in 57m 16s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/7/check/gate-grenade-dsvm/915b281/ : SUCCESS in 24m 24s\n- gate-nova-tox-db-functional http://logs.openstack.org/18/234218/7/check/gate-nova-tox-db-functional/a9df4c6/ : SUCCESS in 11m 39s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/7/check/gate-grenade-dsvm-multinode/5091f33/ : SUCCESS in 50m 52s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/d736e93/ : SUCCESS in 36m 11s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-cells/c14f89b/ : SUCCESS in 33m 37s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/3cfec68/ : SUCCESS in 48m 23s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/7/check/gate-tempest-dsvm-multinode-full/8d49540/ : SUCCESS in 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cbe5163460e7ddde5ded5aedcbcd98d2f546a12d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-08 12:37:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/7/ : SUCCESS in 1h 37m 34s","accounts_in_message":[],"_revision_number":7},{"id":"a5f59a7bdb76d5dae2658ab92e55674c5b528258","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-08 14:18:05.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-234218-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"d9b2401cfacfae25f929d53c412518a0bb2a0cd3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-10 14:30:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/234218/7/3774 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"00c90326e80590bc38e612324e0b72d859be8bc2","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2016-04-11 07:39:21.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"15eb6a3bf41b4c8ef567c40d973dd2b94667bc94","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-04-11 19:09:55.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ae5976e2e0b73f037897ce92758371e5270b962f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-11 19:29:50.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6d32cc83896b4583ad6fb8fb9eee221b/index.html : SUCCESS in 14m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f1935281b5264feeac46eca918cd4dac/index.html : SUCCESS in 15m 35s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7cfa0d102e144963a4ea8084f2406539/index.html : SUCCESS in 15m 30s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/8/check/real-db-upgrade_nova_percona_user_001:th-percona/952c7733db5a4a0183d440544092e45b/index.html : SUCCESS in 16m 39s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"4e15ed6bf3819ea305e1b72387fdc0597c9d028b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-11 19:39:01.000000000","message":"Patch Set 8: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/8/11885/index.html : FAILURE in 27 m 28 s","accounts_in_message":[],"_revision_number":8},{"id":"bbd5f6c0a8e77b500aec67284522c346aff59e52","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-11 22:08:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/8/check/dsvm-tempest-xen/c1c8d39 : SUCCESS in 1h 24m 57s","accounts_in_message":[],"_revision_number":8},{"id":"b68dd27752e79f8eccd77620345d5ca1c6463dac","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-12 00:34:55.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/234218/8/check/check-dsvm-tempest-vz7-exe-minimal/83677ba : FAILURE in 36m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"69071168718cc0ec0840cef0c1e13cb99d4bdf81","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-04-12 08:08:17.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"112f8d0ae3fdf32b40d997283ac9eaca91f6c1cd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-12 08:27:27.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5dbe60707f404b0db1192f2e9910f674/index.html : SUCCESS in 12m 37s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/24254d6317a84d9ab471145ca7f58be3/index.html : SUCCESS in 16m 08s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e9f4ed3a3684486dbf2896069dba9bea/index.html : SUCCESS in 15m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/9/check/real-db-upgrade_nova_percona_user_001:th-percona/3cb1e88a93a9459082c216f0fcbaf662/index.html : SUCCESS in 14m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"dd566833242ad21d262b553d36cd8ad924dc58ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-12 09:14:51.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/9/check/check-ibm-tempest-dsvm-full/094f712/ : SUCCESS in 1h 03m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f4b66a6d4b2571b3b7e9a52bef1c3e4d9c9d2e8f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-12 09:16:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/9/check/gate-nova-docs/580d624//doc/build/html/ : SUCCESS in 3m 07s\n- gate-nova-pep8 http://logs.openstack.org/18/234218/9/check/gate-nova-pep8/81a7afe/ : SUCCESS in 3m 55s\n- gate-nova-python27-db http://logs.openstack.org/18/234218/9/check/gate-nova-python27-db/46a2879/ : SUCCESS in 8m 47s\n- gate-nova-python34-db http://logs.openstack.org/18/234218/9/check/gate-nova-python34-db/021467e/ : SUCCESS in 10m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-full/4027e58/ : SUCCESS in 43m 20s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-postgres-full/641ccf2/ : SUCCESS in 33m 23s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-neutron-full/977dbbc/ : SUCCESS in 1h 06m 02s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/9/check/gate-grenade-dsvm/4b76f01/ : SUCCESS in 41m 40s\n- gate-nova-tox-db-functional http://logs.openstack.org/18/234218/9/check/gate-nova-tox-db-functional/c9714a2/ : SUCCESS in 10m 30s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/9/check/gate-grenade-dsvm-multinode/5cb2756/ : SUCCESS in 41m 07s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/28ad8ec/ : SUCCESS in 38m 53s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-cells/fe8d55d/ : SUCCESS in 27m 23s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/032548b/ : SUCCESS in 45m 14s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/9/check/gate-tempest-dsvm-multinode-full/556bfc4/ : FAILURE in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"33d1784d94694e956e99188d70aff6d112078166","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-12 09:20:13.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/234218/9/check/check-dsvm-tempest-vz7-exe-minimal/f32f572 : FAILURE in 1h 09m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"d01a16eeababb62b729deaa44c8862298fc6255b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-12 09:20:57.000000000","message":"Patch Set 9: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/9/11932/index.html : SUCCESS in 01 h 10 m 13 s","accounts_in_message":[],"_revision_number":9},{"id":"03d2df7eec1705d15260fbe832fe296dcf80bedf","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-12 09:40:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/9/check/dsvm-tempest-xen/497c0df : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":9},{"id":"e338827ffbfd0e8df889f3f9df83d6d00518efcd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-12 09:55:58.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-234218-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"bb3a7b7c95ba19b61c171173b9faf5dad971b345","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-12 09:58:50.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/9/ : SUCCESS in 1h 49m 34s","accounts_in_message":[],"_revision_number":9},{"id":"251a41d01c5824435a1cc7074cccb6547308e1b2","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2016-04-12 11:43:15.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"1c2f4198d5df75b0dfc286531d8308a48198cd64","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-12 15:29:09.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/234218/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"550cd5f41bd4f15e1f6b5c6a2798c444eb092e5a","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-04-12 15:57:26.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c35e210b3461a6fe4efada47d439b8ee583aae06","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-12 17:34:14.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/234218/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"c6cda90e81a9409cb320a03dbba1bf559d0266ef","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-04-13 07:38:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/18/234218/9/check-nova/Nova-ML2-Sriov/a4b5d5e : SUCCESS in 45m 46s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/18/234218/9/check-nova/Nova-MACVTAP-ML2-Sriov/2640ad7 : SUCCESS in 44m 52s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"8398cb88e4c279d5db72056b0ec511d24e1342aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-13 07:56:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/234218/9/3874 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"032efe35eb30377c0e45a6c822136b151ea86f3c","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-04-14 12:39:37.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"53689ac3668142fcbb51c50095b1bc04393b9590","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-14 12:55:52.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0a60a9acb41440af908310af522dc1fe/index.html : SUCCESS in 9m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/38be5f4b58c949839bb4f098a5422473/index.html : SUCCESS in 12m 47s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4a93da1d4421497d907b1e2240630991/index.html : SUCCESS in 10m 44s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/10/check/real-db-upgrade_nova_percona_user_001:th-percona/9e7d3e3917b447bf950fe1aa43a8c459/index.html : SUCCESS in 10m 07s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"04b3114ad24698804e3a9d035252fcca4b11a241","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-14 12:59:48.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/234218/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"129915c2c4040f6ee650b01a0c09124e93bd00fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-14 13:43:41.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/10/check/check-ibm-tempest-dsvm-full/3480e5c/ : SUCCESS in 1h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b4ba2d7561f2e220a8dbe16ee4314bda9f15cbad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-14 13:46:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/10/check/gate-nova-docs/9b31e20//doc/build/html/ : SUCCESS in 5m 06s\n- gate-nova-pep8 http://logs.openstack.org/18/234218/10/check/gate-nova-pep8/f772be1/ : SUCCESS in 4m 41s\n- gate-nova-python27-db http://logs.openstack.org/18/234218/10/check/gate-nova-python27-db/541927a/ : SUCCESS in 13m 13s\n- gate-nova-python34-db http://logs.openstack.org/18/234218/10/check/gate-nova-python34-db/a5f2efb/ : SUCCESS in 11m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-full/051c7d5/ : SUCCESS in 45m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-postgres-full/617a7e3/ : SUCCESS in 47m 40s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-neutron-full/6c6eddf/ : SUCCESS in 54m 45s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/10/check/gate-grenade-dsvm/e2a9947/ : SUCCESS in 47m 23s\n- gate-nova-tox-db-functional http://logs.openstack.org/18/234218/10/check/gate-nova-tox-db-functional/77089c2/ : SUCCESS in 11m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/10/check/gate-grenade-dsvm-multinode/e65e131/ : SUCCESS in 35m 56s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/12bd51b/ : SUCCESS in 33m 10s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-cells/5c4d961/ : SUCCESS in 44m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/9ade7ef/ : SUCCESS in 44m 29s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-multinode-full/f990d52/ : FAILURE in 51m 44s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/18/234218/10/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/cb5f375/ : FAILURE in 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a11eb3b71495e17cd152b8f2756757aafa8f5501","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-14 14:01:17.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/10/check/dsvm-tempest-xen/82b02e0 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":10},{"id":"526efda5ed5a84c3be607547583835c0ec37cff5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-14 14:50:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/234218/10/check/check-dsvm-tempest-vz7-exe-minimal/a356527 : SUCCESS in 1h 07m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"bf6c34d2e27f37b2bbf21b1a9372cd2441cf6bb5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-04-14 14:55:16.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/18/234218/10/check-nova/Nova-ML2-Sriov/638a661 : SUCCESS in 46m 14s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/18/234218/10/check-nova/Nova-MACVTAP-ML2-Sriov/7b937db : SUCCESS in 1h 23m 16s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"372dec24efbe488d79b8c657864317ec3a4c169e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-14 15:21:54.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/10/ : FAILURE in 2h 40m 52s","accounts_in_message":[],"_revision_number":10},{"id":"f2eec06995260872186176ccef57a617af7d656b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-14 15:55:14.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-234218-10 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":10},{"id":"1e19ee7573d2dedf98d7b068750044de467339cc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-15 19:23:10.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/234218/10/3971 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"8cf33e3ea7eb40aaafaf7de5c56f99e1dc2c3146","author":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"date":"2016-05-06 09:50:31.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"cba14ccde3822150967b88318bd16c593326201e","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-06-14 08:27:37.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"4b0ce58e4593c92f66b4baa0ed08628719bdde9b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-14 08:50:27.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/33667ac9270c42ada736c62a7c05e4b3/index.html : SUCCESS in 16m 21s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/8ceb14b9210f4dcb9f7732130c215916/index.html : SUCCESS in 19m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/234218/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/3efc19022d5f4fc6b144dd05a7f1a57b/index.html : SUCCESS in 19m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/234218/11/check/real-db-upgrade_nova_percona_user_001:th-percona/c30e0f3e130c48f390e2f7d512909670/index.html : SUCCESS in 21m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"f5d4b16921ff0152e8ef54c2eb9955b2b217f427","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-14 09:12:38.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/pci/234218/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"6f87548a937f57ad8e965ed8f0e4040e60506098","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-14 09:27:58.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/234218/11/check/check-ibm-tempest-dsvm-full/00ab3d6/ : SUCCESS in 59m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"595f1c5a0854966a1d56f98f6c59ee38cdc7b9b2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-14 09:30:36.000000000","message":"Patch Set 11: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/234218/11/19073/index.html : SUCCESS in 01 h 00 m 42 s","accounts_in_message":[],"_revision_number":11},{"id":"066e087c5fe52e1597c8d3994324e663ece4ec8e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 09:59:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/11/check/gate-nova-docs/488ed41//doc/build/html/ : SUCCESS in 4m 06s\n- gate-nova-pep8 http://logs.openstack.org/18/234218/11/check/gate-nova-pep8/9af4165/ : SUCCESS in 6m 53s\n- gate-nova-python27-db http://logs.openstack.org/18/234218/11/check/gate-nova-python27-db/a681582/ : SUCCESS in 15m 30s\n- gate-nova-python34-db http://logs.openstack.org/18/234218/11/check/gate-nova-python34-db/9688a8d/ : SUCCESS in 12m 45s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-full/066304f/ : SUCCESS in 50m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-postgres-full/eb14adb/ : SUCCESS in 1h 02m 08s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-neutron-full/87dabaa/ : SUCCESS in 1h 09m 32s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/11/check/gate-grenade-dsvm/ddd1f3f/ : SUCCESS in 42m 28s\n- gate-nova-tox-db-functional http://logs.openstack.org/18/234218/11/check/gate-nova-tox-db-functional/a00bc3e/ : SUCCESS in 15m 26s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/11/check/gate-grenade-dsvm-multinode/e47c33f/ : SUCCESS in 47m 22s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b9a018c/ : SUCCESS in 34m 02s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-cells/bd0e61d/ : SUCCESS in 36m 26s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-full-devstack-plugin-ceph/a75c57a/ : SUCCESS in 48m 58s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-multinode-full/afd3862/ : SUCCESS in 55m 42s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/18/234218/11/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/0a680ef/ : SUCCESS in 1h 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cfd26ede73d68869df61da14d5f3efa2ab1b8cff","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-14 10:00:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/234218/11/check/dsvm-tempest-xen/0893a5a : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":11},{"id":"a24973188acd912063be1cd711ac0b7554a8ef88","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-14 10:57:40.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/18/234218/11/check-nova/Nova-ML2-Sriov/c12db3f : SUCCESS in 29m 17s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/18/234218/11/check-nova/Nova-MACVTAP-ML2-Sriov/99c907e : SUCCESS in 32m 52s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"f76aafa54f2fabf9a6aaa5b34f8a5c07e4f9a1e3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-14 10:58:55.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/234218/11/ : SUCCESS in 2h 30m 43s","accounts_in_message":[],"_revision_number":11},{"id":"e1716997cf002ebbff25ac7f2226af9e2529a52b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-06-14 12:32:15.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-234218-11 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":11},{"id":"6d9608d314594ec86aae02c79ac198140417132e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-14 13:08:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/18/234218/11/check/tempest-dsvm-full-nfv/b1233d6 : SUCCESS in 1h 32m 30s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/18/234218/11/check/tempest-dsvm-intel-nfv/748dc29 : SUCCESS in 35m 56s\n- tempest-dsvm-ovsdpdk-nfv http://intel-openstack-ci-logs.ovh/18/234218/11/check/tempest-dsvm-ovsdpdk-nfv/13558ef : SUCCESS in 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ea8489816a29a5fb85bdee168c6d76839b0b0dea","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-07-05 08:32:09.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"011d0446dd9453a3bdab2d7cc99a7b04ea19ee7e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-07-07 00:01:55.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"464a9e50d3aad0a85119860c52541a6028c13123","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 00:04:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":11},{"id":"1647dbeaa7c0c83c8e020b5620baccefc1b7da2b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 01:37:51.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/18/234218/11/gate/gate-nova-docs/a67475b//doc/build/html/ : SUCCESS in 2m 54s\n- gate-nova-pep8 http://logs.openstack.org/18/234218/11/gate/gate-nova-pep8/6002c2d/ : SUCCESS in 4m 24s\n- gate-nova-python27-db http://logs.openstack.org/18/234218/11/gate/gate-nova-python27-db/576d85e/ : SUCCESS in 9m 53s\n- gate-nova-python34-db http://logs.openstack.org/18/234218/11/gate/gate-nova-python34-db/585d670/ : SUCCESS in 9m 15s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/234218/11/gate/gate-tempest-dsvm-full/0220216/ : SUCCESS in 34m 19s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/234218/11/gate/gate-tempest-dsvm-postgres-full/5b23c75/ : SUCCESS in 44m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/234218/11/gate/gate-tempest-dsvm-neutron-full/29b40f7/ : SUCCESS in 1h 02m 58s\n- gate-grenade-dsvm http://logs.openstack.org/18/234218/11/gate/gate-grenade-dsvm/08ca560/ : SUCCESS in 35m 21s\n- gate-nova-tox-db-functional http://logs.openstack.org/18/234218/11/gate/gate-nova-tox-db-functional/0d51439/ : SUCCESS in 13m 33s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/234218/11/gate/gate-grenade-dsvm-multinode/16f9246/ : SUCCESS in 48m 29s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/234218/11/gate/gate-tempest-dsvm-cells/983f3e5/ : SUCCESS in 31m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/18/234218/11/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/6bdcadf/ : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":11},{"id":"c8ea01f8c32c9309f337a18535cd1068f4df6fb1","date":"2016-07-07 01:37:53.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"29ec93d99d71f4abd689d97c0448a85c7e24c809","revisions":{"4755d87c08ab19754313779d2cd84b491874a7cf":{"kind":"REWORK","_number":1,"created":"2015-10-13 13:38:35.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/1"}}},"commit":{"parents":[{"commit":"dc5b751c2c7125946a5455522523ab0e38ec3688","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc5b751c2c7125946a5455522523ab0e38ec3688"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"subject":"Use instance objection instead of db record","message":"Use instance objection instead of db record\n\nmake console manager use object of instance\ninstead of db record in the add_console method\nalso, updated unit test of related usage from\ndb record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4755d87c08ab19754313779d2cd84b491874a7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4755d87c08ab19754313779d2cd84b491874a7cf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dc5b751c2c7125946a5455522523ab0e38ec3688","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b501dd4a6e51e0fef4a633ce4712a63c0e360073":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2015-11-24 10:51:21.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/2"}}},"commit":{"parents":[{"commit":"2106373a118afc1253c9ab0edb8f62117f3d6d2a","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2106373a118afc1253c9ab0edb8f62117f3d6d2a"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-11-24 10:51:04.000000000","tz":0},"subject":"Use instance objection instead of db record","message":"Use instance objection instead of db record\n\nmake console manager use object of instance\ninstead of db record in the add_console method\nalso, updated unit test of related usage from\ndb record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b501dd4a6e51e0fef4a633ce4712a63c0e360073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b501dd4a6e51e0fef4a633ce4712a63c0e360073"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2106373a118afc1253c9ab0edb8f62117f3d6d2a","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"87137de7421b64d1e2ac259b4bde38cfe27182ef":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2015-11-25 12:18:42.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/3"}}},"commit":{"parents":[{"commit":"b9770ed62801c9f67a0944625485de4b49ca3035","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b9770ed62801c9f67a0944625485de4b49ca3035"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-11-21 13:33:27.000000000","tz":480},"subject":"Use instance objection instead of db record","message":"Use instance objection instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87137de7421b64d1e2ac259b4bde38cfe27182ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87137de7421b64d1e2ac259b4bde38cfe27182ef"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b9770ed62801c9f67a0944625485de4b49ca3035","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2c0a0cf29275f2ffcc0af3cedb965731770d0bb8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2015-12-15 13:34:57.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/4"}}},"commit":{"parents":[{"commit":"4328df48350e265f32fd8e9775bf98ec094c5be9","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4328df48350e265f32fd8e9775bf98ec094c5be9"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-12-15 13:34:50.000000000","tz":0},"subject":"Use instance objection instead of db record","message":"Use instance objection instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c0a0cf29275f2ffcc0af3cedb965731770d0bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c0a0cf29275f2ffcc0af3cedb965731770d0bb8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4328df48350e265f32fd8e9775bf98ec094c5be9","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a0648f0d796c56a88cf0dd7431a802b808047934":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2015-12-18 15:22:02.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/5"}}},"commit":{"parents":[{"commit":"ce890c0a467dc10135853712637637bd4fecfbbd","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce890c0a467dc10135853712637637bd4fecfbbd"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-12-18 00:25:06.000000000","tz":480},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0648f0d796c56a88cf0dd7431a802b808047934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0648f0d796c56a88cf0dd7431a802b808047934"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ce890c0a467dc10135853712637637bd4fecfbbd","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"395c55df7ae3a0e317dbdba7edc52d7fd0a08133":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2016-03-07 12:10:30.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/6"}}},"commit":{"parents":[{"commit":"56870b92187e4957fa748bb6414b35fb72bb5e35","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56870b92187e4957fa748bb6414b35fb72bb5e35"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2016-03-07 12:10:26.000000000","tz":0},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/395c55df7ae3a0e317dbdba7edc52d7fd0a08133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/395c55df7ae3a0e317dbdba7edc52d7fd0a08133"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"56870b92187e4957fa748bb6414b35fb72bb5e35","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0f04bea57594d8f44cdd047eb0db2ac22867d7f9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2016-04-07 19:01:30.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/7"}}},"commit":{"parents":[{"commit":"353f79558e9732009536e6d7562220e7bffbd0e5","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/353f79558e9732009536e6d7562220e7bffbd0e5"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2016-03-20 14:24:48.000000000","tz":480},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f04bea57594d8f44cdd047eb0db2ac22867d7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f04bea57594d8f44cdd047eb0db2ac22867d7f9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"353f79558e9732009536e6d7562220e7bffbd0e5","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fdb85b78b91e98d1658bafd1b7ca7f57c8fb9afa":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2016-04-11 19:09:55.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/8"}}},"commit":{"parents":[{"commit":"6836482474a8e890c0b9e0c3613530f339d4e6c6","subject":"Remove stub_compute config options for console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6836482474a8e890c0b9e0c3613530f339d4e6c6"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2016-03-21 10:07:54.000000000","tz":480},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdb85b78b91e98d1658bafd1b7ca7f57c8fb9afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdb85b78b91e98d1658bafd1b7ca7f57c8fb9afa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6836482474a8e890c0b9e0c3613530f339d4e6c6","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0a3863ee88b3c2d95a0af0c640963b69d8bcdaa2":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2016-04-12 08:08:17.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/9"}}},"commit":{"parents":[{"commit":"c0612cae63df03ba13f9954b5025dfc9f95dddcb","subject":"Deprecate stub_compute config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0612cae63df03ba13f9954b5025dfc9f95dddcb"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2016-03-21 11:02:21.000000000","tz":480},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a3863ee88b3c2d95a0af0c640963b69d8bcdaa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a3863ee88b3c2d95a0af0c640963b69d8bcdaa2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c0612cae63df03ba13f9954b5025dfc9f95dddcb","is_merged_in_target_branch":false,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cd6e9de224e61f53d2e35561bfcae2218153230e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2016-04-14 12:39:37.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/10"}}},"commit":{"parents":[{"commit":"3d0272f9fe3d18c18f0cca732ee02321797d2e93","subject":"Remove stub_compute config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d0272f9fe3d18c18f0cca732ee02321797d2e93"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2016-03-22 11:20:25.000000000","tz":480},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd6e9de224e61f53d2e35561bfcae2218153230e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd6e9de224e61f53d2e35561bfcae2218153230e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3d0272f9fe3d18c18f0cca732ee02321797d2e93","is_merged_in_target_branch":true,"change_id":"If38148e178e50a348810312826b16cbd34ba550b","change_number":234217,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"29ec93d99d71f4abd689d97c0448a85c7e24c809":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2016-06-14 08:27:37.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/18/234218/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/234218/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/234218/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/234218/11"}}},"commit":{"parents":[{"commit":"05cab2666f947c5a337b530701aae877968e834b","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05cab2666f947c5a337b530701aae877968e834b"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2015-10-07 03:41:18.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2016-06-14 08:27:35.000000000","tz":0},"subject":"Use instance object instead of db record","message":"Use instance object instead of db record\n\nmake console manager use object of instance instead of db\nrecord in the add_console method also, updated unit test\nof related usage from db record to object.\n\nChange-Id: I7ff21e35f584bde2ece6ccda71f1fa93ff2cfff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/29ec93d99d71f4abd689d97c0448a85c7e24c809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/29ec93d99d71f4abd689d97c0448a85c7e24c809"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"05cab2666f947c5a337b530701aae877968e834b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
