)]}'
{"id":"openstack%2Fnova~252059","triplet_id":"openstack%2Fnova~master~I337d9ce3836b0a4629b5cc77d85f7578c894f360","project":"openstack/nova","branch":"master","topic":"bug/1521675","hashtags":[],"change_id":"I337d9ce3836b0a4629b5cc77d85f7578c894f360","subject":"Pre-load meta and sys_meta when constructing InstanceMetadata","status":"ABANDONED","created":"2015-12-01 20:52:13.000000000","updated":"2016-02-11 18:06:10.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2011990be676fe2908022e128c1c8a4cdf9976a","_number":252059,"virtual_id_number":252059,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"date":"2016-02-10 12:45:01.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-02-11 05:47:30.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-02-10 12:52:24.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-02-10 11:03:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2016-02-10 10:20:24.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2016-02-10 10:58:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"date":"2016-02-10 18:56:59.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2016-02-10 11:32:43.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-02-10 20:33:06.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"date":"2016-02-10 10:12:59.000000000","_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2016-02-10 11:27:23.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"date":"2016-02-11 02:30:30.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-12-20 21:50:20.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2015-12-20 22:06:49.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-12-23 14:06:36.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2015-12-23 15:56:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 10:12:59.000000000","updated_by":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"reviewer":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"state":"REVIEWER"},{"updated":"2016-02-10 10:20:24.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 10:58:54.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 11:03:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 11:27:23.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 11:32:43.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 12:45:01.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 12:52:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 18:56:59.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-10 20:33:06.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-11 02:30:30.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-11 05:47:30.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"0817faa8db9a56503fb3f7b9d681a673faf5ff58","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-12-01 20:52:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a627659fee1e928aab503065f646235cd37fd432","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-01 21:13:15.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8442cafaf7b2423db66614d567eb6806/index.html : SUCCESS in 13m 08s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/db09db0e647a4b8dab6abb93964c0719/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b0f1e827f0cb4dec8a515635cc1e7c0b/index.html : SUCCESS in 12m 26s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/1/check/real-db-upgrade_nova_percona_user_001:th-percona/4cb15a85f84348549733a54e6b2bf11a/index.html : SUCCESS in 13m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/1b575ff1d9564db0b26023d5e46dc4f2/index.html : SUCCESS in 15m 02s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/1/check/real-db-upgrade_nova_percona_user_002:th-percona/b658ced7d38640cf98447bd30dabf41f/index.html : SUCCESS in 15m 21s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"f8641e4f347ce41810509f2a4a62b87e52f974e2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-01 21:51:26.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/1/check/check-ibm-tempest-dsvm-full/d7fc7f8/ : SUCCESS in 58m 31s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/1/check/check-ibm-tempest-dsvm-postgres-full/e77a2f3/ : SUCCESS in 54m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ad8ce0964dbe9bff21416913660f5158f1e7dea","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-01 21:57:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/252059/1/ : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":1},{"id":"317ec86886aa034308f7bdc2daef780db2e0cfda","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-01 21:59:47.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/59/252059/1/17676/index.html : SUCCESS in 01 h 03 m 40 s","accounts_in_message":[],"_revision_number":1},{"id":"52e7ef3d342e00a6ee1934732fdbe2e70ecfde45","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-01 22:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/59/252059/1/check/gate-nova-pep8/9302166/ : SUCCESS in 9m 25s\n- gate-nova-docs http://docs-draft.openstack.org/59/252059/1/check/gate-nova-docs/f4dff62//doc/build/html/ : SUCCESS in 14m 16s\n- gate-nova-python27 http://logs.openstack.org/59/252059/1/check/gate-nova-python27/bedefbe/ : SUCCESS in 11m 49s\n- gate-nova-python34 http://logs.openstack.org/59/252059/1/check/gate-nova-python34/16c9cf7/ : SUCCESS in 13m 09s\n- gate-tempest-dsvm-full http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-full/f8e37b4/ : SUCCESS in 42m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-postgres-full/72fb096/ : SUCCESS in 44m 12s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-neutron-full/5e2d757/ : SUCCESS in 57m 44s\n- gate-grenade-dsvm http://logs.openstack.org/59/252059/1/check/gate-grenade-dsvm/d7a2ba1/ : SUCCESS in 54m 22s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-large-ops/582e71f/ : SUCCESS in 42m 07s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-neutron-large-ops/d5144cc/ : SUCCESS in 33m 59s\n- gate-nova-tox-functional http://logs.openstack.org/59/252059/1/check/gate-nova-tox-functional/3b3ed45/ : SUCCESS in 15m 56s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/59/252059/1/check/gate-grenade-dsvm-multinode/361f41a/ : SUCCESS in 1h 04m 18s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3a19e2f/ : SUCCESS in 50m 23s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-nova-v20-api/d90942e/ : SUCCESS in 44m 10s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-nova-v20-api-legacy/068d578/ : SUCCESS in 44m 54s\n- gate-tempest-dsvm-cells http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-cells/c96ab5e/ : SUCCESS in 53m 49s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-full-ceph/ac66179/ : SUCCESS in 49m 50s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/59/252059/1/check/gate-tempest-dsvm-multinode-full/e1615c9/ : FAILURE in 59m 40s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/59/252059/1/check/gate-compute-api-guide/1e7ad6a//api-guide/build/html/ : SUCCESS in 6m 15s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/59/252059/1/check/gate-nova-python27-constraints/74cef5f/ : SUCCESS in 14m 29s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/59/252059/1/check/gate-nova-python34-constraints/11cc6c0/ : SUCCESS in 12m 39s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/59/252059/1/check/gate-nova-docs-constraints/1e2a84c//doc/build/html/ : SUCCESS in 15m 08s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/59/252059/1/check/gate-nova-pep8-constraints/6f480cf/ : SUCCESS in 9m 02s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/59/252059/1/check/gate-nova-tox-constraints-functional-constraints/dfb21cc/ : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba744381fe4c30b2f73e823be0891c6681956909","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-01 22:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/59/252059/1/check/dsvm-tempest-xen/465b80a : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"933abe6e3a0d4d8f37029babd45915e86dd4d8aa","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-01 23:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/59/252059/1/check/check-dsvm-vz7-tempest-exe-minimal/e73306a/ : SUCCESS in 54m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"e341eba0d62c86666588b1b0cd1cab63cef843c3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-02 00:11:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/59/252059/1/check/dsvm-tempest-cinder/3080f70/ : SUCCESS in 47m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"62c2c546b739f0384dff99aad53f8f72e6d02171","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-02 10:37:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/252059/1/Nova-ML2-Sriov : ABORTED in 17m 22s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"2990aed62cd0e32f22b52386724ae91b27df84ed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-02 23:42:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/252059/1 \n\n* nova-dsvm-tempest http://208.91.1.172/logs/252059/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"57f9568fb4f70489cf210c8091ca3409341d77ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 21:32:43.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"63686558df49ae3f7450cf88f2f793c213b7511d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-12-11 03:03:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35f4498259172e2cc7708bd26ea42d4306328c82","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-11 03:03:59.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/252059/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cb2c2eebf07bf37ec9b58bd5c567000c23852640","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-11 03:27:31.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/614e1562ef2941df919ebc5011a7d113/index.html : SUCCESS in 14m 38s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d2a10c64684f40af84b3423dd5413ff1/index.html : SUCCESS in 14m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/63222acaf9b14012ae62e8429abca5b2/index.html : SUCCESS in 15m 51s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/2/check/real-db-upgrade_nova_percona_user_001:th-percona/9cd890e5a4fe4669930b7a6e0be5b635/index.html : SUCCESS in 16m 28s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a2202a397f834699a423c860a1471c61/index.html : SUCCESS in 12m 43s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/2/check/real-db-upgrade_nova_percona_user_002:th-percona/94c92fa49c96467d9ce25d8dde2301a3/index.html : SUCCESS in 14m 57s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"52bb554a3ee5d20c70350d3ae11659783f555cf5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-11 03:52:50.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/2/check/check-ibm-tempest-dsvm-full/81125a4/ : SUCCESS in 48m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2868fc81a65588c9ad1c9105d2383d39515507d0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-11 04:01:07.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/59/252059/2/19145/index.html : SUCCESS in 55 m 04 s","accounts_in_message":[],"_revision_number":2},{"id":"607d7b82b3cf084a7a5a7a78a824385c382ee5f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-11 04:10:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/59/252059/2/check/gate-nova-pep8/d4eebc2/ : SUCCESS in 10m 10s\n- gate-nova-docs http://docs-draft.openstack.org/59/252059/2/check/gate-nova-docs/a20ab8d//doc/build/html/ : SUCCESS in 14m 34s\n- gate-nova-python27 http://logs.openstack.org/59/252059/2/check/gate-nova-python27/723cc10/ : SUCCESS in 14m 03s\n- gate-nova-python34 http://logs.openstack.org/59/252059/2/check/gate-nova-python34/4976fe6/ : SUCCESS in 13m 10s\n- gate-tempest-dsvm-full http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-full/155bd28/ : SUCCESS in 53m 12s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-postgres-full/243293a/ : SUCCESS in 50m 53s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-neutron-full/efe6414/ : SUCCESS in 1h 00m 59s\n- gate-grenade-dsvm http://logs.openstack.org/59/252059/2/check/gate-grenade-dsvm/01d81d7/ : SUCCESS in 40m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-large-ops/a079607/ : SUCCESS in 29m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-neutron-large-ops/fd57a97/ : SUCCESS in 29m 10s\n- gate-nova-tox-functional http://logs.openstack.org/59/252059/2/check/gate-nova-tox-functional/002f60c/ : SUCCESS in 16m 51s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/59/252059/2/check/gate-grenade-dsvm-multinode/f633cf3/ : SUCCESS in 51m 10s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ced5e00/ : SUCCESS in 44m 42s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-nova-v20-api/ecafc85/ : SUCCESS in 40m 29s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-nova-v20-api-legacy/aed970c/ : SUCCESS in 40m 40s\n- gate-tempest-dsvm-cells http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-cells/c8da4aa/ : SUCCESS in 41m 04s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-full-ceph/6c7b92a/ : SUCCESS in 53m 20s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/59/252059/2/check/gate-tempest-dsvm-multinode-full/968fdbd/ : SUCCESS in 51m 00s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/59/252059/2/check/gate-compute-api-guide/ccf9097//api-guide/build/html/ : SUCCESS in 8m 42s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/59/252059/2/check/gate-nova-python27-constraints/7b08421/ : SUCCESS in 15m 11s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/59/252059/2/check/gate-nova-python34-constraints/f6044e9/ : SUCCESS in 12m 51s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/59/252059/2/check/gate-nova-docs-constraints/f5a97d5//doc/build/html/ : SUCCESS in 12m 58s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/59/252059/2/check/gate-nova-pep8-constraints/a6b0905/ : SUCCESS in 7m 35s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/59/252059/2/check/gate-nova-tox-constraints-functional-constraints/3494b58/ : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18e2e34ff68bdc7e257187b796a9be59b3a0aa58","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-11 04:19:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/59/252059/2/check/dsvm-tempest-xen/7a915f4 : FAILURE in 1h 11m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7392e386239d1b3da9eb3d04d4568fb187f8e3b0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-11 04:30:13.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/252059/2/ : FAILURE in 1h 25m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f453a0188e26bc681247c21083bd66e5fa8b862a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-11 06:06:56.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/252059/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a383c826e4c4d3241376319e2bf373ac4e1ab3f2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-12-11 09:10:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd4047e735ffb98d90458c771cf488c4f09b5081","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-11 17:31:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- check-dsvm-vz7-tempest-exe-minimal http://openstack-3rd-party-ci-logs.odin.net/59/252059/2/check/check-dsvm-vz7-tempest-exe-minimal/56aa466/ : FAILURE in 1h 02m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027Virtuozzo CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"eadd3fff0979aff835276704924687c7c211efde","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-11 19:25:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cinder http://openstack-3rd-party-ci-logs.odin.net/59/252059/2/check/dsvm-tempest-cinder/6c8e461/ : FAILURE in 1h 47m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"8ab71f422d54cf4d9069bcb743e5735f164bbbd9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-12 04:55:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/252059/2/Nova-ML2-Sriov : SUCCESS in 1h 05m 54s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"80fbde41f0e82d9b229acb359c443d26161b0f13","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-12 08:43:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/252059/2 \n\n* nova-dsvm-vc6-tempest http://208.91.1.172/logs/252059/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"89dfee8a1a6e51f0c8e6119a09a95fcba4276af0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2015-12-16 21:08:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7fd45776681f2e3b6abefc9b712894cf6b8bbe1","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-12-20 20:41:50.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"74e5727d75548057333e529b7ac079ada184d723","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-12-20 20:56:44.000000000","message":"Patch Set 3: Code-Review+1\n\nRebased to get a dependent patch in. The dependent patch should take care of the concern raised by sahid.","accounts_in_message":[],"_revision_number":3},{"id":"507957434614f85789babf65170cbe5613d0416e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2015-12-20 21:01:13.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/252059/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2beb3de9f04086d93bf7433bee2a158a32079b30","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2015-12-20 21:14:28.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/cdf01cdb07af40f59a0b80a64f02e6c4/index.html : SUCCESS in 14m 40s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/391ffc5406ae44aaacce842f24c5cff3/index.html : SUCCESS in 13m 58s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/250b40357e3245e6838fdc1a42686dee/index.html : SUCCESS in 15m 27s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/3/check/real-db-upgrade_nova_percona_user_001:th-percona/6f7a127a4a9e459796dae61752c4c38f/index.html : SUCCESS in 17m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d9bb3782911e4572a055d4e066184ed6/index.html : SUCCESS in 15m 33s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/3/check/real-db-upgrade_nova_percona_user_002:th-percona/96e387e0f57548a4b25fca5bda81b84f/index.html : SUCCESS in 15m 31s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"b45ecd20e294bcc194c0da68ffcf10545e25f6f2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-20 21:15:45.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/3/check/check-ibm-tempest-dsvm-full/43eb446/ : FAILURE in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c48d68e0dd39abdac858b68a6cdc9f233890245","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2015-12-20 21:30:59.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/59/252059/3/check/check-dsvm-tempest-vz7-exe-minimal/5aef2f4 : FAILURE in 46m 22s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"b8ed748f717dba3194364aeee1c4a98868c47326","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2015-12-20 21:44:09.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/59/252059/3/20528/index.html : SUCCESS in 59 m 32 s","accounts_in_message":[],"_revision_number":3},{"id":"93c2e8da2f1a23ea6ecc967cfe60796f97aa7341","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-20 21:49:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/59/252059/3/check/gate-nova-pep8/14d469d/ : SUCCESS in 8m 56s\n- gate-nova-docs http://docs-draft.openstack.org/59/252059/3/check/gate-nova-docs/0a661bf//doc/build/html/ : SUCCESS in 13m 46s\n- gate-nova-python27 http://logs.openstack.org/59/252059/3/check/gate-nova-python27/fe3abe8/ : SUCCESS in 14m 35s\n- gate-nova-python34 http://logs.openstack.org/59/252059/3/check/gate-nova-python34/c41b565/ : SUCCESS in 13m 36s\n- gate-tempest-dsvm-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-full/9ff4f5c/ : FAILURE in 1h 06m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-postgres-full/043aa76/ : SUCCESS in 39m 28s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-neutron-full/a48976d/ : SUCCESS in 52m 46s\n- gate-grenade-dsvm http://logs.openstack.org/59/252059/3/check/gate-grenade-dsvm/49ca836/ : SUCCESS in 49m 54s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-large-ops/8fc3944/ : SUCCESS in 29m 21s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-neutron-large-ops/8ed5a1a/ : SUCCESS in 36m 06s\n- gate-nova-tox-functional http://logs.openstack.org/59/252059/3/check/gate-nova-tox-functional/ec7063a/ : SUCCESS in 15m 40s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/59/252059/3/check/gate-grenade-dsvm-multinode/d247a96/ : SUCCESS in 53m 48s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/bbad11b/ : SUCCESS in 45m 14s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-nova-v20-api/c710bba/ : SUCCESS in 40m 09s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-nova-v20-api-legacy/8ab056d/ : SUCCESS in 35m 39s\n- gate-tempest-dsvm-cells http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-cells/2f904a7/ : SUCCESS in 45m 22s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-full-ceph/7fe1914/ : SUCCESS in 46m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/298c6fd/ : SUCCESS in 47m 33s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-multinode-full/11998e4/ : SUCCESS in 51m 50s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/59/252059/3/check/gate-compute-api-guide/b2c9281//api-guide/build/html/ : SUCCESS in 8m 36s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/59/252059/3/check/gate-nova-commit-message/209cab1/ : FAILURE in 1m 23s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-python27-constraints/c0fe321/ : SUCCESS in 14m 36s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-python34-constraints/159b757/ : SUCCESS in 9m 25s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/59/252059/3/check/gate-nova-docs-constraints/9ad83b0//doc/build/html/ : SUCCESS in 14m 44s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-pep8-constraints/905f160/ : SUCCESS in 9m 22s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-tox-constraints-functional-constraints/28117e5/ : SUCCESS in 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f17383d5332be82e14081e38b66cd8cc8b3ed65","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2015-12-20 21:50:20.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1526737 https://bugs.launchpad.net/bugs/1355573\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"dca902ded88a7bde72957d60726f9dd1ad8ca895","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2015-12-20 21:56:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/59/252059/3/check/dsvm-tempest-xen/a47ec0a : FAILURE in 1h 08m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8f66140b488b7f37af63064bfb1a45e1765dd7bc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2015-12-20 22:06:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/59/252059/3/check/dsvm-tempest-kvm/1307532 : SUCCESS in 41m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"8aed3942c74671c41335bb1cc908df85f1f776df","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2015-12-20 22:12:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/252059/3/Nova-ML2-Sriov : SUCCESS in 57m 34s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"b328ccf4ec8b8cc8a51cb088e8f42401c729fb0b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-21 00:04:43.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/252059/3/ : FAILURE in 3h 21m 15s","accounts_in_message":[],"_revision_number":3},{"id":"42b9293561bd825be7fdd3700d31f1c6a814dcc9","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2015-12-21 09:32:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4ecdda87733738ea3d677bb8030e6959d091af01","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-12-21 09:46:22.000000000","message":"Patch Set 3: -Code-Review\n\nThanks Hans your help on this serie seems to be useful.","accounts_in_message":[],"_revision_number":3},{"id":"74f6a439d9d7d3847e30d7890ebd052957605de4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2015-12-21 09:58:02.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/252059/3/ : FAILURE in 25m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9530653317bfc3db40b9cfc1b34e52e292bf2261","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2015-12-21 10:02:15.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/3/check/check-ibm-tempest-dsvm-full/c7c7fbb/ : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7e189ad8868121f3a607fee79de08916149c2484","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-21 10:33:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/59/252059/3/check/gate-nova-pep8/5a3635a/ : SUCCESS in 9m 38s\n- gate-nova-docs http://docs-draft.openstack.org/59/252059/3/check/gate-nova-docs/d50fef8//doc/build/html/ : SUCCESS in 14m 02s\n- gate-nova-python27 http://logs.openstack.org/59/252059/3/check/gate-nova-python27/bbae0e0/ : SUCCESS in 11m 27s\n- gate-nova-python34 http://logs.openstack.org/59/252059/3/check/gate-nova-python34/49cde4b/ : SUCCESS in 13m 18s\n- gate-tempest-dsvm-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-full/55682b2/ : SUCCESS in 59m 20s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-postgres-full/fa61b06/ : SUCCESS in 53m 48s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-neutron-full/c9a50e5/ : SUCCESS in 56m 56s\n- gate-grenade-dsvm http://logs.openstack.org/59/252059/3/check/gate-grenade-dsvm/09459b5/ : SUCCESS in 57m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-large-ops/6ff5470/ : SUCCESS in 36m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-neutron-large-ops/44eb650/ : SUCCESS in 34m 04s\n- gate-nova-tox-functional http://logs.openstack.org/59/252059/3/check/gate-nova-tox-functional/d117b1e/ : SUCCESS in 18m 11s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/59/252059/3/check/gate-grenade-dsvm-multinode/79cf734/ : SUCCESS in 50m 26s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f71c816/ : FAILURE in 52m 06s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-nova-v20-api/505f4a2/ : SUCCESS in 50m 15s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-nova-v20-api-legacy/22d6126/ : SUCCESS in 47m 02s\n- gate-tempest-dsvm-cells http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-cells/bbec28d/ : SUCCESS in 47m 48s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-full-ceph/ec041ae/ : SUCCESS in 50m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/c6f298c/ : SUCCESS in 52m 18s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/59/252059/3/check/gate-tempest-dsvm-multinode-full/dfd9964/ : FAILURE in 55m 03s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/59/252059/3/check/gate-compute-api-guide/8e5057a//api-guide/build/html/ : SUCCESS in 8m 36s (non-voting)\n- gate-nova-commit-message http://logs.openstack.org/59/252059/3/check/gate-nova-commit-message/7d0b1f7/ : FAILURE in 1m 50s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-python27-constraints/f8c67d1/ : SUCCESS in 16m 37s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-python34-constraints/3f3e57b/ : SUCCESS in 13m 28s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/59/252059/3/check/gate-nova-docs-constraints/da4c28e//doc/build/html/ : SUCCESS in 15m 06s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-pep8-constraints/e248ba3/ : SUCCESS in 7m 26s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/59/252059/3/check/gate-nova-tox-constraints-functional-constraints/af3ae30/ : SUCCESS in 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"39ffb176379f1890fa0e96e67208ef71dd0dfcfb","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2015-12-23 14:06:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"eb4abc41dc224766a87018ecca7bd0bdf99dc82f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2015-12-23 15:56:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/252059/3 \n\nhttp://p3-ci-int.eng.vmware.com/job/nova-vc6-dsvm-tempest/453/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"8242c3a899154cacb15ea66e4aa78f1c80f3a48f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-15 01:20:11.000000000","message":"Patch Set 3:\n\nxen: recheck","accounts_in_message":[],"_revision_number":3},{"id":"296904c088ec7009f77fd9c4b3c8aa89c2dac5a1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-15 02:54:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/59/252059/3/check/dsvm-tempest-xen/b5e3e89 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c7799cd451ea41a102d7701261053ae159a23328","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 23:23:21.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"db22dec961f39a186686df7fbb8982c8f8cd8e47","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2016-02-10 09:09:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2ab597e6edf3f4286555b147711ff8f39b6a906f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-10 09:27:38.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/252059/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"5074ebe06e1b9d491b9fb69d6cba21afd91ea926","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-10 09:34:36.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/820b99983ade417aa617e9e38a6fb72a/index.html : SUCCESS in 12m 42s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/84fe0df87da14b8896e48462eb08c42c/index.html : SUCCESS in 15m 21s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/61d9faa78a804f82aaaedfbbe5a3f3e2/index.html : SUCCESS in 11m 23s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/4/check/real-db-upgrade_nova_percona_user_001:th-percona/9f0e7a26784c4d6daee2da4deae3c9c0/index.html : SUCCESS in 15m 43s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a9f86404c6c547bc97f9792699e448df/index.html : SUCCESS in 12m 47s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/4/check/real-db-upgrade_nova_percona_user_002:th-percona/09ace9efd47e4ed5b5cea0aefe0e2d30/index.html : SUCCESS in 14m 06s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"e5a021cabc75d40c7df2d5b2db6dfc8c5d6b1ce4","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2016-02-10 10:06:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"34af63a517e41c08cd7ca6a860bf20675d52826a","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2016-02-10 10:12:59.000000000","message":"Patch Set 5:\n\nNew unit tests merged in Change-Id: I77fa013939a4394690b81b74a887ede66f29c6c1 similar to the ones that was added in this change. Modify those instead of adding almost identical ones here since they needed to be changed anyway to pass unit tests.","accounts_in_message":[],"_revision_number":5},{"id":"1a20ef6cbb621844b2250b4751a68c9fb81bbfb7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-10 10:20:24.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/252059/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"7bd352338f3ffeab4051eea33967b3d69e1aec67","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-10 10:58:54.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/108fa38d4d68496c97521eafb994b4ef/index.html : SUCCESS in 18m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/207b020398e94857995a37842b1f70ef/index.html : SUCCESS in 17m 36s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1325a023af104cc09e6445a8c3a44dd6/index.html : SUCCESS in 15m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/5/check/real-db-upgrade_nova_percona_user_001:th-percona/323729c0f3724d0184cd60d1dfb134a9/index.html : SUCCESS in 15m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/59/252059/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/335a168b80564f4ca8cea25b5dc9ca7d/index.html : SUCCESS in 12m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/59/252059/5/check/real-db-upgrade_nova_percona_user_002:th-percona/a8c1269392a94c6ba6454185c5bd8414/index.html : SUCCESS in 19m 00s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"06d8fc0e2b12c03ce2986edb6b74e1dafcfec2f5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-10 11:03:42.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/5/check/check-ibm-tempest-dsvm-full/08978dc/ : FAILURE in 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8e278dacf9eaea86a94ee976751b46c3c779e142","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-10 11:27:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/59/252059/5/check/dsvm-tempest-xen/06121a7 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":5},{"id":"8be743ffbc1b18119eaff2dd9f275125243e7d80","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-10 11:32:43.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/59/252059/5/5812/index.html : SUCCESS in 01 h 18 m 34 s","accounts_in_message":[],"_revision_number":5},{"id":"b5574062a8d0accc6de5d73546e0129a322ba3eb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-10 11:40:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/59/252059/5/check/gate-nova-pep8/ad79d2b/ : SUCCESS in 2m 15s\n- gate-nova-docs http://docs-draft.openstack.org/59/252059/5/check/gate-nova-docs/01f1b45//doc/build/html/ : SUCCESS in 7m 43s\n- gate-nova-python27 http://logs.openstack.org/59/252059/5/check/gate-nova-python27/a98eaac/ : SUCCESS in 8m 45s\n- gate-nova-python34 http://logs.openstack.org/59/252059/5/check/gate-nova-python34/b2e958b/ : SUCCESS in 27m 17s\n- gate-tempest-dsvm-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-full/6fd7776/ : SUCCESS in 1h 19m 55s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-postgres-full/d03c80e/ : FAILURE in 1h 19m 40s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-neutron-full/2ab82a8/ : SUCCESS in 48m 18s\n- gate-grenade-dsvm http://logs.openstack.org/59/252059/5/check/gate-grenade-dsvm/d051673/ : SUCCESS in 34m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-large-ops/f7cccb4/ : SUCCESS in 26m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-neutron-large-ops/6333cb3/ : SUCCESS in 25m 19s\n- gate-nova-tox-functional http://logs.openstack.org/59/252059/5/check/gate-nova-tox-functional/cda9fed/ : SUCCESS in 10m 53s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/59/252059/5/check/gate-grenade-dsvm-multinode/7f16e48/ : FAILURE in 40m 12s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/deeefe9/ : SUCCESS in 40m 10s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-nova-v20-api/cc3b799/ : SUCCESS in 30m 27s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-nova-v20-api-legacy/9c768b6/ : SUCCESS in 34m 36s\n- gate-tempest-dsvm-cells http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-cells/6c676ac/ : SUCCESS in 36m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-full-ceph/a799f2d/ : SUCCESS in 47m 09s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/b1f969e/ : SUCCESS in 47m 11s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-multinode-full/8ec2910/ : SUCCESS in 1h 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d8a6b876df70ff388b7f702bec567a17e667dfa","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-10 12:45:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/59/252059/5/check/check-dsvm-tempest-vz7-exe-minimal/5b968b3 : SUCCESS in 1h 12m 03s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"447f77079db29db7b12214c82c090e461a0b3c55","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-10 12:52:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/252059/5/Nova-ML2-Sriov : SUCCESS in 38m 59s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"55f63517f61ce271f57ac98e79fe06ba8601e028","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-10 15:07:07.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/252059/5/ : FAILURE in 3h 29m 28s","accounts_in_message":[],"_revision_number":5},{"id":"357c2bc987ea14c1ae68aec170bd5f96398c21a5","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2016-02-10 15:43:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1c6222fbab64be4792262067e0c2fd817972246b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-10 16:07:54.000000000","message":"Patch Set 5:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/59/252059/5 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"64358723ad1b0bd9d1e8ec949d7b8459ef15584c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-10 16:40:46.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/59/252059/5/check/check-ibm-tempest-dsvm-full/fc9f44c/ : SUCCESS in 56m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d1f087c0d9a417712cca46606a0ce919e9ead381","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-10 18:25:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/59/252059/5/check/gate-nova-pep8/c194b11/ : SUCCESS in 2m 28s\n- gate-nova-docs http://docs-draft.openstack.org/59/252059/5/check/gate-nova-docs/1f2f96d//doc/build/html/ : SUCCESS in 7m 25s\n- gate-nova-python27 http://logs.openstack.org/59/252059/5/check/gate-nova-python27/6a7439d/ : SUCCESS in 12m 29s\n- gate-nova-python34 http://logs.openstack.org/59/252059/5/check/gate-nova-python34/ec8c47d/ : SUCCESS in 28m 41s\n- gate-tempest-dsvm-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-full/2551117/ : SUCCESS in 37m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-postgres-full/e53e6df/ : SUCCESS in 32m 26s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-neutron-full/965cc39/ : SUCCESS in 48m 15s\n- gate-grenade-dsvm http://logs.openstack.org/59/252059/5/check/gate-grenade-dsvm/f26afb1/ : SUCCESS in 32m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-large-ops/6389284/ : SUCCESS in 22m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-neutron-large-ops/045d7b5/ : SUCCESS in 21m 58s\n- gate-nova-tox-functional http://logs.openstack.org/59/252059/5/check/gate-nova-tox-functional/f6ca73f/ : SUCCESS in 12m 18s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/59/252059/5/check/gate-grenade-dsvm-multinode/c2a1a92/ : SUCCESS in 38m 52s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/87396c7/ : SUCCESS in 30m 25s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-cells/0cafa87/ : SUCCESS in 26m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-full-ceph/4323d64/ : SUCCESS in 37m 45s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/51eabec/ : SUCCESS in 40m 36s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/59/252059/5/check/gate-tempest-dsvm-multinode-full/c6bb4d4/ : FAILURE in 1h 06m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6a86b6ca4b3e3e2579d5814f2046ac46fac4f8bd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-10 18:56:59.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-59-252059-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"b929688b8e6e534c3d90bebd5e5d6adaf48f42bb","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-10 20:33:06.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/252059/5/ : FAILURE in 1h 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"c8c65d79b2cc00bfa99943e6a213d331b882ad63","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-11 02:30:30.000000000","message":"Patch Set 5:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/59/252059/5 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"f4356dd94e42ff4bbfa9cc71020c2ca39a569012","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-11 05:47:30.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"5ba868a8d0cda4ac7aedafe4c7c77800b74b6975","author":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"date":"2016-02-11 07:22:18.000000000","message":"Patch Set 5:\n\nIdentical fix Change-Id: I8e565b6ce976fcbe6194b5960a3ebf2b76a45416 merged ahead of this one.","accounts_in_message":[],"_revision_number":5},{"id":"ec887d16a678a20627de79009a6fd2ff744e651b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-02-11 18:06:10.000000000","message":"Abandoned\n\nDuped by https://review.openstack.org/#/c/276861/ - thanks for pointing it out Hans.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f448d5987481e79681ef00705df1af58570ce6e5","revisions":{"9f2aa4d89a69e4219d608dff6143dee7da13ff7c":{"kind":"REWORK","_number":1,"created":"2015-12-01 20:52:13.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/59/252059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/252059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/252059/1"}}},"commit":{"parents":[{"commit":"ae70557c14f482f61ad02cac4f5a144855fdaa8c","subject":"objects: lazy-load instance.security_groups more efficiently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae70557c14f482f61ad02cac4f5a144855fdaa8c"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-01 20:41:59.000000000","tz":-480},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-01 20:52:01.000000000","tz":-480},"subject":"Pre-load meta/sys_meta/sec_groups when constructing InstanceMetadata","message":"Pre-load meta/sys_meta/sec_groups when constructing InstanceMetadata\n\nThe InstanceMetadata.__init__ method is getting the security group list\ndirectly, and lazy-loading metadata (for launch_metadata) and\nsystem_metadata (when getting the password).\n\nWe can make this more efficient by joining those tables in the initial\ninstance get query that happens in get_metadata_by_instance_id, which\nmeans fewer round trips to conductor.\n\nAdds a new unit test since the existing tests go through the handler\nobject and mock out the base.get_metadata_by_instance_id method.\n\nCloses-Bug: #1521675\n\nChange-Id: I337d9ce3836b0a4629b5cc77d85f7578c894f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f2aa4d89a69e4219d608dff6143dee7da13ff7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f2aa4d89a69e4219d608dff6143dee7da13ff7c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ae70557c14f482f61ad02cac4f5a144855fdaa8c","is_merged_in_target_branch":false,"change_id":"Ie578967bf22288766b3acc2f7b4f55735e228b6f","change_number":251965,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1f3d9d91f1827974c91863b3420aaf022a815c92":{"kind":"REWORK","_number":2,"created":"2015-12-11 03:03:18.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/59/252059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/252059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/252059/2"}}},"commit":{"parents":[{"commit":"bb94c815fdf0b47f934d97109cd215eafcb85b2a","subject":"objects: lazy-load instance.security_groups more efficiently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb94c815fdf0b47f934d97109cd215eafcb85b2a"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-11 02:08:27.000000000","tz":-480},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-11 03:00:37.000000000","tz":-480},"subject":"Pre-load meta and sys_meta when constructing InstanceMetadata","message":"Pre-load meta and sys_meta when constructing InstanceMetadata\n\nThe InstanceMetadata.__init__ method is lazy-loading metadata\n(for launch_metadata) and system_metadata (when getting the password).\n\nWe can make this more efficient by joining those tables in the initial\ninstance get query that happens in get_metadata_by_instance_id, which\nmeans fewer round trips to conductor.\n\nAdds a new unit test since the existing tests go through the handler\nobject and mock out the base.get_metadata_by_instance_id method.\n\nCloses-Bug: #1521675\n\nChange-Id: I337d9ce3836b0a4629b5cc77d85f7578c894f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f3d9d91f1827974c91863b3420aaf022a815c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f3d9d91f1827974c91863b3420aaf022a815c92"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bb94c815fdf0b47f934d97109cd215eafcb85b2a","is_merged_in_target_branch":false,"change_id":"Ie578967bf22288766b3acc2f7b4f55735e228b6f","change_number":251965,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e13df517014df51d81b3798305a77645bbc3c5e2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2015-12-20 20:41:50.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/59/252059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/252059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/252059/3"}}},"commit":{"parents":[{"commit":"cdd483aafd8255c6adf28d78dc9ef60c30c8551b","subject":"objects: lazy-load instance.security_groups more efficiently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdd483aafd8255c6adf28d78dc9ef60c30c8551b"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-11 02:08:27.000000000","tz":-480},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2015-12-20 20:32:12.000000000","tz":60},"subject":"Pre-load meta and sys_meta when constructing InstanceMetadata","message":"Pre-load meta and sys_meta when constructing InstanceMetadata\n\nThe InstanceMetadata.__init__ method is lazy-loading metadata\n(for launch_metadata) and system_metadata (when getting the password).\n\nWe can make this more efficient by joining those tables in the initial\ninstance get query that happens in get_metadata_by_instance_id, which\nmeans fewer round trips to conductor.\n\nAdds a new unit test since the existing tests go through the handler\nobject and mock out the base.get_metadata_by_instance_id method.\n\nCloses-Bug: #1521675\n\nChange-Id: I337d9ce3836b0a4629b5cc77d85f7578c894f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e13df517014df51d81b3798305a77645bbc3c5e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e13df517014df51d81b3798305a77645bbc3c5e2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cdd483aafd8255c6adf28d78dc9ef60c30c8551b","is_merged_in_target_branch":true,"change_id":"Ie578967bf22288766b3acc2f7b4f55735e228b6f","change_number":251965,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"239b93fef93261e69f03eeae90328bbe06437b5a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-02-10 09:09:38.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/59/252059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/252059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/252059/4"}}},"commit":{"parents":[{"commit":"3a6c1ca78c5da579d994fa1dba53705e089e8d63","subject":"Merge \"Config options: centralize options in conductor api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a6c1ca78c5da579d994fa1dba53705e089e8d63"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-11 02:08:27.000000000","tz":-480},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2016-02-10 08:57:44.000000000","tz":60},"subject":"Pre-load meta and sys_meta when constructing InstanceMetadata","message":"Pre-load meta and sys_meta when constructing InstanceMetadata\n\nThe InstanceMetadata.__init__ method is lazy-loading metadata\n(for launch_metadata) and system_metadata (when getting the password).\n\nWe can make this more efficient by joining those tables in the initial\ninstance get query that happens in get_metadata_by_instance_id, which\nmeans fewer round trips to conductor.\n\nAdds a new unit test since the existing tests go through the handler\nobject and mock out the base.get_metadata_by_instance_id method.\n\nCloses-Bug: #1521675\n\nChange-Id: I337d9ce3836b0a4629b5cc77d85f7578c894f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/239b93fef93261e69f03eeae90328bbe06437b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/239b93fef93261e69f03eeae90328bbe06437b5a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3a6c1ca78c5da579d994fa1dba53705e089e8d63","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f448d5987481e79681ef00705df1af58570ce6e5":{"kind":"REWORK","_number":5,"created":"2016-02-10 10:06:29.000000000","uploader":{"_account_id":6450,"name":"Hans Lindgren","email":"hanlind@kth.se","username":"hanlind"},"ref":"refs/changes/59/252059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/252059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/252059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/252059/5"}}},"commit":{"parents":[{"commit":"3a6c1ca78c5da579d994fa1dba53705e089e8d63","subject":"Merge \"Config options: centralize options in conductor api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a6c1ca78c5da579d994fa1dba53705e089e8d63"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2015-12-11 02:08:27.000000000","tz":-480},"committer":{"name":"Hans Lindgren","email":"hanlind@kth.se","date":"2016-02-10 09:57:23.000000000","tz":60},"subject":"Pre-load meta and sys_meta when constructing InstanceMetadata","message":"Pre-load meta and sys_meta when constructing InstanceMetadata\n\nThe InstanceMetadata.__init__ method is lazy-loading metadata\n(for launch_metadata) and system_metadata (when getting the password).\n\nWe can make this more efficient by joining those tables in the initial\ninstance get query that happens in get_metadata_by_instance_id, which\nmeans fewer round trips to conductor.\n\nCloses-Bug: #1521675\n\nChange-Id: I337d9ce3836b0a4629b5cc77d85f7578c894f360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f448d5987481e79681ef00705df1af58570ce6e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f448d5987481e79681ef00705df1af58570ce6e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3a6c1ca78c5da579d994fa1dba53705e089e8d63","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
