)]}'
{"id":"openstack%2Fnova~271878","triplet_id":"openstack%2Fnova~master~Ia93f63a60f57c660baaf6e11ec4fb8824545b675","project":"openstack/nova","branch":"master","topic":"pep8-e251","hashtags":[],"change_id":"Ia93f63a60f57c660baaf6e11ec4fb8824545b675","subject":"Enable pep8 rule E121","status":"ABANDONED","created":"2016-01-25 03:11:54.000000000","updated":"2016-02-04 07:12:57.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af24762b98de97f6ba4234b3ef603b764b10fd2c","_number":271878,"virtual_id_number":271878,"owner":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2016-02-02 08:09:32.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-02-02 07:45:50.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-02-02 07:24:43.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-02-02 07:37:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2016-02-02 07:09:04.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":1,"date":"2016-02-02 07:27:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":1,"date":"2016-02-04 06:10:51.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-02-02 09:26:37.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-02-02 06:56:13.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"date":"2016-02-02 12:54:44.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-02-02 06:39:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":2,"date":"2016-02-02 06:39:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"all":[{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-02-02 17:24:30.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-01-25 17:27:42.000000000","updated_by":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"reviewer":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"state":"REVIEWER"},{"updated":"2016-01-25 21:15:51.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2016-01-26 16:40:38.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2016-02-02 02:07:30.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 06:56:13.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 07:09:04.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 07:24:43.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 07:27:02.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 07:37:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 07:45:50.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-02-02 08:09:32.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 09:26:37.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 12:54:44.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-04 06:10:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72b720e37e5dc528ac25a3d942a3ce0f7d20f0bc","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-01-25 03:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e761446005c58132720a35851073580e295e8a2","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-01-25 03:25:47.000000000","message":"Patch Set 1:\n\nHere are the counts of the other pep8 TODOs in case anyone is curious: http://paste.openstack.org/show/484787/","accounts_in_message":[],"_revision_number":1},{"id":"4b908111d7cb03fe33dfed6d9583fe5d7f219098","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-25 03:55:29.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/78/271878/1/check/check-dsvm-tempest-vz7-exe-minimal/917ec84 : FAILURE in 42m 54s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"e0b019e16c512d415415486c61b44437086a140b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-25 04:07:25.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/78/271878/1/check/check-ibm-tempest-dsvm-full/4800a42/ : SUCCESS in 54m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14351502ef421a9799c28c3c6472f8f21b376826","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-25 04:17:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/78/271878/1/check/gate-nova-pep8/49c09a2/ : SUCCESS in 8m 10s\n- gate-nova-docs http://docs-draft.openstack.org/78/271878/1/check/gate-nova-docs/13b3da0//doc/build/html/ : SUCCESS in 10m 06s\n- gate-nova-python27 http://logs.openstack.org/78/271878/1/check/gate-nova-python27/6702121/ : SUCCESS in 11m 50s\n- gate-nova-python34 http://logs.openstack.org/78/271878/1/check/gate-nova-python34/2248a66/ : SUCCESS in 28m 50s\n- gate-tempest-dsvm-full http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-full/d5c152f/ : SUCCESS in 53m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-postgres-full/3e69ee1/ : SUCCESS in 50m 28s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-neutron-full/8470268/ : SUCCESS in 54m 46s\n- gate-grenade-dsvm http://logs.openstack.org/78/271878/1/check/gate-grenade-dsvm/670bdfa/ : SUCCESS in 46m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-large-ops/3fc2a7b/ : SUCCESS in 40m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-neutron-large-ops/afd8505/ : SUCCESS in 38m 30s\n- gate-nova-releasenotes http://logs.openstack.org/78/271878/1/check/gate-nova-releasenotes/822ff69/ : FAILURE in 30m 32s\n- gate-nova-tox-functional http://logs.openstack.org/78/271878/1/check/gate-nova-tox-functional/bb21b21/ : SUCCESS in 19m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/78/271878/1/check/gate-grenade-dsvm-multinode/b3f0756/ : SUCCESS in 57m 37s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f31ee37/ : FAILURE in 42m 46s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-nova-v20-api/ad9fb99/ : SUCCESS in 35m 06s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-nova-v20-api-legacy/239c726/ : SUCCESS in 36m 08s\n- gate-tempest-dsvm-cells http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-cells/d306f49/ : SUCCESS in 40m 19s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-full-ceph/ac9e923/ : SUCCESS in 45m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/27dea92/ : SUCCESS in 49m 15s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/78/271878/1/check/gate-tempest-dsvm-multinode-full/afb68c7/ : SUCCESS in 1h 02m 19s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/78/271878/1/check/gate-compute-api-guide/610b029//api-guide/build/html/ : SUCCESS in 7m 26s (non-voting)\n- gate-nova-python27-constraints http://logs.openstack.org/78/271878/1/check/gate-nova-python27-constraints/fdd38a2/ : SUCCESS in 12m 17s (non-voting)\n- gate-nova-python34-constraints http://logs.openstack.org/78/271878/1/check/gate-nova-python34-constraints/0aa7707/ : SUCCESS in 26m 02s (non-voting)\n- gate-nova-docs-constraints http://docs-draft.openstack.org/78/271878/1/check/gate-nova-docs-constraints/09df8f3//doc/build/html/ : SUCCESS in 13m 10s (non-voting)\n- gate-nova-pep8-constraints http://logs.openstack.org/78/271878/1/check/gate-nova-pep8-constraints/5c93f15/ : SUCCESS in 11m 28s (non-voting)\n- gate-nova-tox-constraints-functional-constraints http://logs.openstack.org/78/271878/1/check/gate-nova-tox-constraints-functional-constraints/49aa88e/ : SUCCESS in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a428ff2b45d90937004533db1d221c1025d5624","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-25 04:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/271878/1/Nova-ML2-Sriov : SUCCESS in 31m 25s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"f162c70e11c0cad4edff196c1d18b257d7d7dc58","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-25 04:30:00.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/271878/1/4089/index.html : SUCCESS in 01 h 15 m 04 s","accounts_in_message":[],"_revision_number":1},{"id":"8e0c44a2f69c19f4e22e527dcb43ca7590b7ff1c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-25 04:33:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/78/271878/1/check/dsvm-tempest-xen/d0f41b3 : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a609092a1573dc711bf48e827a8e6ee212071724","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-25 04:52:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/271878/1/ : SUCCESS in 1h 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"583e229b4086b1ee064dccb9fab70df36adac47b","author":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"date":"2016-01-25 17:27:42.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI have to say that this is just on the edge of reviewability, and it\u0027ll probably be a bit of a nightmare to keep up with merge conflicts (it appears that there are 47 conflicting reviews).  Good luck getting it merged :)","accounts_in_message":[],"_revision_number":1},{"id":"911a9a29fdcf996a3905aa06ad8b0faea588d20b","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-01-25 20:00:05.000000000","message":"Patch Set 1:\n\n@Kevin:\n\nIt is indeed on the edge of reviewability -- thanks for playing along ;) When I started this, I figured it would be a much smaller diff with a bunch of one line whitespace changes.\n\nPS. I\u0027ll remove those backslashes during the next inevitable merge conflict.\n\nThanks!\n\n--diana","accounts_in_message":[],"_revision_number":1},{"id":"78b810e7ce69616f1433355433b09300d5d66e55","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-01-25 21:15:51.000000000","message":"Patch Set 1: Code-Review-1\n\nThe E12* rules are kind of turned off on purpose. They are things which are overly pedantic, and don\u0027t add a ton of readability most of the time. \n\nIf we are going to change things like this, we should do it during first milestone, as this current conflicts with 47 in flight patches, so can wreck functional code we are trying to get in.","accounts_in_message":[],"_revision_number":1},{"id":"79b1998d9476fdc4e8e38b0d419fe8c2324229c9","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-01-25 21:29:05.000000000","message":"Patch Set 1:\n\nOkie dokie :)\n\nThe tox.ini file said that they were TODOs, so I figured it was fair game. Perhaps we should change this then: \"The rest of the ignores are TODOs\".\n\nhttps://github.com/openstack/nova/blob/54f79255768f85330cb85b6e9c80be271cf06cb8/tox.ini#L150\n\nCheers,\n\n--diana","accounts_in_message":[],"_revision_number":1},{"id":"54be56e6c320af291c8ecb08d07c5ba7f47efe0b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-26 09:21:06.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/271878/1/1679 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8f4347f4613ea5bd4ede9567df0ab8b3fdb923f4","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2016-01-26 16:40:38.000000000","message":"Patch Set 1: Code-Review+2\n\nI am ok with this going in, but not until Newton opens because of the number of other patches it conflicts with. So, let\u0027s not merge this right now.","accounts_in_message":[],"_revision_number":1},{"id":"4038193ec831d315ee13d165edd6219c04991d07","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2016-01-26 16:40:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6ea78f8dcb379997fec90892c0f9ce9fb660c06c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-29 00:27:18.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4d9eec055ec44dfc84487faf81726102/index.html : SUCCESS in 14m 26s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a712d6fe941f462bbf22cbbafa89fe55/index.html : SUCCESS in 12m 27s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fc66869f2278412a807c5ad8ab4ac722/index.html : SUCCESS in 17m 51s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/1/check/real-db-upgrade_nova_percona_user_001:th-percona/0eff9af1ba38486e91e2c2c3ab3341d5/index.html : SUCCESS in 15m 09s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c8ca00be986940468ad4fd464b2c651e/index.html : SUCCESS in 13m 46s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/1/check/real-db-upgrade_nova_percona_user_002:th-percona/f4d880995e284612a20bac9bac20b8e9/index.html : SUCCESS in 13m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"5e93f973fbd3e8f0301ec8cd38022bc6ba29b1f1","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-02-02 00:54:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c1cfa0efb70b69f828139a628b04a8e0bede2c31","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 01:05:36.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/78/271878/2/check/check-dsvm-tempest-vz7-exe-minimal/9ecb6da : FAILURE in 9m 08s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"550abe4f0c24103206ad9186c55c00a1dfdc36e7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-02 01:22:54.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/271878/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"20111ddce5595185dae760f851f12ce4ab2d8126","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-02 01:36:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/271878/2/Nova-ML2-Sriov : FAILURE in 40m 45s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a540df74807a8d0a0f679fdc6e080f74198044a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-02 01:51:22.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/78/271878/2/check/check-ibm-tempest-dsvm-full/44203e8/ : SUCCESS in 55m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1b722761bb279b9b7e53b07acb1311c9680e260","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-02 02:07:30.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/271878/2/4861/index.html : SUCCESS in 01 h 09 m 45 s","accounts_in_message":[],"_revision_number":2},{"id":"302aa209e064dfee48a5e067da9201fe1558cd69","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 02:07:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/78/271878/2/check/gate-nova-pep8/1786256/ : FAILURE in 8m 42s\n- gate-nova-docs http://docs-draft.openstack.org/78/271878/2/check/gate-nova-docs/be0393c//doc/build/html/ : SUCCESS in 12m 29s\n- gate-nova-python27 http://logs.openstack.org/78/271878/2/check/gate-nova-python27/3cb1397/ : SUCCESS in 16m 12s\n- gate-nova-python34 http://logs.openstack.org/78/271878/2/check/gate-nova-python34/0e2b2ab/ : SUCCESS in 27m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-full/d5ef8f9/ : SUCCESS in 52m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-postgres-full/12b8e4e/ : SUCCESS in 40m 12s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-neutron-full/adcef2d/ : SUCCESS in 1h 00m 02s\n- gate-grenade-dsvm http://logs.openstack.org/78/271878/2/check/gate-grenade-dsvm/7f49355/ : SUCCESS in 46m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-large-ops/f34c460/ : SUCCESS in 27m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-neutron-large-ops/f7b23e9/ : SUCCESS in 32m 16s\n- gate-nova-releasenotes http://docs-draft.openstack.org/78/271878/2/check/gate-nova-releasenotes/a8a2297//releasenotes/build/html/ : SUCCESS in 7m 37s\n- gate-nova-tox-functional http://logs.openstack.org/78/271878/2/check/gate-nova-tox-functional/414ad63/ : SUCCESS in 14m 44s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/78/271878/2/check/gate-grenade-dsvm-multinode/fe92ce8/ : SUCCESS in 59m 23s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/45222c1/ : SUCCESS in 1h 02m 31s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-nova-v20-api/5ff751c/ : SUCCESS in 49m 11s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-nova-v20-api-legacy/fadbe57/ : SUCCESS in 35m 44s\n- gate-tempest-dsvm-cells http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-cells/41bd648/ : SUCCESS in 38m 02s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-full-ceph/0f1d4ac/ : SUCCESS in 1h 09m 01s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/0fc3d35/ : SUCCESS in 1h 00m 18s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/78/271878/2/check/gate-tempest-dsvm-multinode-full/b8d2f20/ : SUCCESS in 1h 06m 48s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/78/271878/2/check/gate-compute-api-guide/6f3c42b//api-guide/build/html/ : SUCCESS in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"160b360c1126b1161b82654d8658e22a486e184e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-02 02:10:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/78/271878/2/check/dsvm-tempest-xen/0153135 : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":2},{"id":"04e688c2b8663982d5ef65493473337e0b2128de","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-02 02:47:26.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/34fc7181e5e64eb2bea8e32f44c1a9b1/index.html : SUCCESS in 13m 28s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2413beda60f74c74b3577d18656bda9d/index.html : SUCCESS in 22m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ad0fa3c17cf34c86ba133170d8c7ca59/index.html : SUCCESS in 13m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/2/check/real-db-upgrade_nova_percona_user_001:th-percona/4483d345e772499bb4105e57bbf424b8/index.html : SUCCESS in 1h 38m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/5042ef6b192b454ea629ee0f7adf506a/index.html : SUCCESS in 11m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/2/check/real-db-upgrade_nova_percona_user_002:th-percona/4629ac71dd584ecc8d4ee73bacd19fce/index.html : SUCCESS in 20m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"2cd3902dc313c185b162b0d8a59197454166db63","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 04:24:24.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/271878/2/ : FAILURE in 3h 28m 36s","accounts_in_message":[],"_revision_number":2},{"id":"95e5689db0c4a942f4222fb3726042b995df732a","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-02-02 06:22:55.000000000","message":"Patch Set 2: Workflow-1\n\nWIP until Newton opens though because it conflicts with too many other patches.","accounts_in_message":[],"_revision_number":2},{"id":"db4e934b392cce643f1ed26bd4da720442f4929f","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-02-02 06:33:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7579c1e817bfe192b6d7e6e56b1806cdbae530fa","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-02-02 06:39:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"30e7a4ebca6592894923e35f5928d00da6fc4b8e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-02 06:51:24.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/271878/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"7fe201ce582cb26c36fe27a97aedd7ba7ce09058","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-02 06:56:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/78/271878/4/check/dsvm-tempest-xen/d04fad1 : FAILURE in 10m 54s","accounts_in_message":[],"_revision_number":4},{"id":"d9a371a59e304564800667acb179bfe67a080ee5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-02 07:09:04.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/271878/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"2b016dba3cef63ceacae57efbce91843c2fda842","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-02 07:24:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/271878/4/Nova-ML2-Sriov : FAILURE in 40m 15s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"6fa8e52e9ba792c8cd4580d8349aaab6869ac781","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-02 07:27:02.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/69747fb27914454dbdf8e8d42eedd1b2/index.html : SUCCESS in 13m 49s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9e01bd17f0be4ed08bb55188861910fc/index.html : SUCCESS in 15m 27s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a2c51aabbe6843aa868bec83f93c65f0/index.html : SUCCESS in 14m 21s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/4/check/real-db-upgrade_nova_percona_user_001:th-percona/4f7bf6d5f21c48c88cd847205f9b3586/index.html : SUCCESS in 16m 16s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/78/271878/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/34d0369ea23849f3a0dd1c9d07faa010/index.html : SUCCESS in 13m 33s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/78/271878/4/check/real-db-upgrade_nova_percona_user_002:th-percona/cc85a0d23c8740db95fb0e696b9151c6/index.html : SUCCESS in 13m 41s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"9c810a186ae1293da74fd2f6dee44833024b13cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-02 07:37:24.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/78/271878/4/check/check-ibm-tempest-dsvm-full/5424883/ : SUCCESS in 55m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"593ed70d698621328833c39bee3672620c4e4c1e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 07:45:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/78/271878/4/check/gate-nova-pep8/f2b0358/ : FAILURE in 7m 49s\n- gate-nova-docs http://docs-draft.openstack.org/78/271878/4/check/gate-nova-docs/94a2166//doc/build/html/ : SUCCESS in 12m 24s\n- gate-nova-python27 http://logs.openstack.org/78/271878/4/check/gate-nova-python27/9d0598d/ : SUCCESS in 19m 10s\n- gate-nova-python34 http://logs.openstack.org/78/271878/4/check/gate-nova-python34/eb37f9a/ : SUCCESS in 28m 44s\n- gate-tempest-dsvm-full http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-full/2fac2cf/ : FAILURE in 11m 12s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-postgres-full/564e885/ : FAILURE in 10m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-neutron-full/d561ad1/ : SUCCESS in 1h 04m 34s\n- gate-grenade-dsvm http://logs.openstack.org/78/271878/4/check/gate-grenade-dsvm/b83d208/ : FAILURE in 12m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-large-ops/0b1f953/ : FAILURE in 11m 07s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-neutron-large-ops/6d23faf/ : SUCCESS in 36m 04s\n- gate-nova-releasenotes http://docs-draft.openstack.org/78/271878/4/check/gate-nova-releasenotes/9f54b67//releasenotes/build/html/ : SUCCESS in 7m 43s\n- gate-nova-tox-functional http://logs.openstack.org/78/271878/4/check/gate-nova-tox-functional/8f0a46c/ : SUCCESS in 15m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/78/271878/4/check/gate-grenade-dsvm-multinode/b923898/ : FAILURE in 24m 34s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2ee9869/ : SUCCESS in 50m 37s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-nova-v20-api/5cf950a/ : FAILURE in 12m 05s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-nova-v20-api-legacy/8037ec3/ : SUCCESS in 37m 26s\n- gate-tempest-dsvm-cells http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-cells/07eeb84/ : FAILURE in 12m 16s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-full-ceph/80a4042/ : FAILURE in 11m 59s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/580b8c2/ : FAILURE in 13m 54s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/78/271878/4/check/gate-tempest-dsvm-multinode-full/0c5e057/ : FAILURE in 11m 29s (non-voting)\n- gate-compute-api-guide http://docs-draft.openstack.org/78/271878/4/check/gate-compute-api-guide/5f36ad0//api-guide/build/html/ : SUCCESS in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2527d1434ae1256de16d0bc6d2b2288c6a8059df","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 08:09:32.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/78/271878/4/check/check-dsvm-tempest-vz7-exe-minimal/ea55ab4 : FAILURE in 1h 27m 07s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"8e1b0e1f2a07e11477d8ab338e2ed871f31954d5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 09:26:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/271878/4/ : SUCCESS in 2h 45m 33s","accounts_in_message":[],"_revision_number":4},{"id":"cb9e4fb7498ea83ac4c2b978c5f873f489eb34a2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 11:06:10.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/78/271878/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"154d87b44820e8caf80e1fc68696b322ec7c5cea","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 12:54:44.000000000","message":"Patch Set 4:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/78/271878/4 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":4},{"id":"bb33f87c982238e0862c07ac8ac229d2692e7ab9","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-02-02 17:24:30.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"6304d201bd334c5fe9d43950dc0d182d2b36cbd7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-04 06:10:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/271878/4/1969 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"f978064b1efafd2a09ec11ccc328ceb7181487ff","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-02-04 07:12:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ba223dad54b6426cf09dbbfc1ca59c78fc1dc730","revisions":{"fd4eae4413e26c45eadf985e0a8483309c7c8c7d":{"kind":"REWORK","_number":1,"created":"2016-01-25 03:11:54.000000000","uploader":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"ref":"refs/changes/78/271878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/271878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/271878/1"}}},"commit":{"parents":[{"commit":"aa9db1ccef646c07f4946676feeac762ced8997a","subject":"Merge \"doc: add version discovery guideline in api concept doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa9db1ccef646c07f4946676feeac762ced8997a"}]}],"author":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-01-25 02:21:50.000000000","tz":-300},"committer":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-01-25 03:10:38.000000000","tz":-300},"subject":"Enable pep8 rule E121","message":"Enable pep8 rule E121\n\nOf the pep8 TODOs in toxi.ini, E121 is one of the few doable ones\nbecause there are less than 100 of those errors.\n\nE121: continuation line under-indented for hanging indent\n\nChange-Id: Ia93f63a60f57c660baaf6e11ec4fb8824545b675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd4eae4413e26c45eadf985e0a8483309c7c8c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd4eae4413e26c45eadf985e0a8483309c7c8c7d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa9db1ccef646c07f4946676feeac762ced8997a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fbe8f1030b4926be1a6d6ce13a7a38ebb6db53a0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2016-02-02 00:54:27.000000000","uploader":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"ref":"refs/changes/78/271878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/271878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/271878/2"}}},"commit":{"parents":[{"commit":"2fc930a84e0e934c3f3d2c24679c698aeb98f8df","subject":"Merge \"Replace stubs.Set with stub_out (db)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fc930a84e0e934c3f3d2c24679c698aeb98f8df"}]}],"author":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-01-25 02:21:50.000000000","tz":-300},"committer":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-02-02 00:13:36.000000000","tz":-300},"subject":"Enable pep8 rule E121","message":"Enable pep8 rule E121\n\nOf the pep8 TODOs in toxi.ini, E121 is one of the few doable ones\nbecause there are less than 100 of those errors.\n\nE121: continuation line under-indented for hanging indent\n\nChange-Id: Ia93f63a60f57c660baaf6e11ec4fb8824545b675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbe8f1030b4926be1a6d6ce13a7a38ebb6db53a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbe8f1030b4926be1a6d6ce13a7a38ebb6db53a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2fc930a84e0e934c3f3d2c24679c698aeb98f8df","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"439198cf5f65a64695ec51611c08e476c5c2f1f5":{"kind":"REWORK","_number":3,"created":"2016-02-02 06:33:17.000000000","uploader":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"ref":"refs/changes/78/271878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/271878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/271878/3"}}},"commit":{"parents":[{"commit":"0d8a3d90fbd308a05719ae6b14da692399eb09d1","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d8a3d90fbd308a05719ae6b14da692399eb09d1"}]}],"author":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-01-25 02:21:50.000000000","tz":-300},"committer":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-02-02 06:33:00.000000000","tz":-300},"subject":"Enable pep8 rule E121","message":"Enable pep8 rule E121\n\nOf the pep8 TODOs in toxi.ini, E121 is one of the few doable ones\nbecause there are less than 100 of those errors.\n\nE121: continuation line under-indented for hanging indent\n\nChange-Id: Ia93f63a60f57c660baaf6e11ec4fb8824545b675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/439198cf5f65a64695ec51611c08e476c5c2f1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/439198cf5f65a64695ec51611c08e476c5c2f1f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0d8a3d90fbd308a05719ae6b14da692399eb09d1","is_merged_in_target_branch":true,"change_id":"I747855c76ddbe20567753aded6bf08310346af42","change_number":274923,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ba223dad54b6426cf09dbbfc1ca59c78fc1dc730":{"kind":"REWORK","_number":4,"created":"2016-02-02 06:39:54.000000000","uploader":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"ref":"refs/changes/78/271878/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/271878/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/271878/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/271878/4"}}},"commit":{"parents":[{"commit":"0d8a3d90fbd308a05719ae6b14da692399eb09d1","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d8a3d90fbd308a05719ae6b14da692399eb09d1"}]}],"author":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-01-25 02:21:50.000000000","tz":-300},"committer":{"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","date":"2016-02-02 06:37:33.000000000","tz":-300},"subject":"Enable pep8 rule E121","message":"Enable pep8 rule E121\n\nOf the pep8 TODOs in toxi.ini, E121 is one of the few doable ones\nbecause there are less than 100 of those errors.\n\nE121: continuation line under-indented for hanging indent\n\nChange-Id: Ia93f63a60f57c660baaf6e11ec4fb8824545b675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba223dad54b6426cf09dbbfc1ca59c78fc1dc730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba223dad54b6426cf09dbbfc1ca59c78fc1dc730"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0d8a3d90fbd308a05719ae6b14da692399eb09d1","is_merged_in_target_branch":true,"change_id":"I747855c76ddbe20567753aded6bf08310346af42","change_number":274923,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
