)]}'
{"id":"openstack%2Fnova~273042","triplet_id":"openstack%2Fnova~master~I3c83bd68a0e2bbbcdd6d955722dbc9f9fc528113","project":"openstack/nova","branch":"master","topic":"smaller_dhcpbridge","hashtags":[],"change_id":"I3c83bd68a0e2bbbcdd6d955722dbc9f9fc528113","subject":"Spread allocations of fixed ips","status":"MERGED","created":"2016-01-27 13:37:51.000000000","updated":"2016-02-02 11:27:09.000000000","submitted":"2016-02-02 11:27:07.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4672c39a5131071ca43a3b2aaa49813f4325b044","_number":273042,"virtual_id_number":273042,"owner":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-02-02 11:27:07.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},{"value":1,"date":"2016-02-02 01:36:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2016-02-02 02:23:53.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},{"value":-1,"date":"2016-02-02 02:12:03.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"all":[{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2016-02-02 07:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":2,"date":"2016-02-02 10:12:22.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2016-02-02 03:26:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":2,"date":"2016-02-02 10:20:06.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":2,"date":"2016-02-02 11:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":1,"date":"2016-02-02 10:12:22.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-01-28 17:04:26.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2016-01-28 17:23:32.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2016-01-28 18:03:26.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-28 20:04:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-28 20:10:33.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2016-01-29 00:21:36.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-29 01:05:32.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-29 12:04:04.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2016-01-29 21:14:02.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2016-01-31 16:22:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-01 09:29:58.000000000","updated_by":{"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},"reviewer":{"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},"state":"REVIEWER"},{"updated":"2016-02-01 17:55:19.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-01 19:24:22.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 01:36:42.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 02:12:03.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 02:23:53.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-02-02 03:26:37.000000000","updated_by":{"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},"reviewer":{"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},"state":"REVIEWER"},{"updated":"2016-02-02 07:28:04.000000000","updated_by":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"reviewer":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"state":"REVIEWER"},{"updated":"2016-02-02 10:12:22.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-02-02 10:20:06.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2016-02-02 11:18:03.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2016-02-02 11:27:07.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"6a0f07831c7ef06aeb46b623372e98fcc1ba436b","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-01-27 13:37:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f9ebad130c5efaec7034a243cc7c90d4b4649e6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-27 14:19:49.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/273042/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2598ffd35ae2ece2cc518fbc9b80069627b7f407","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-27 14:54:23.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/42/273042/1/4397/index.html : SUCCESS in 01 h 13 m 02 s","accounts_in_message":[],"_revision_number":1},{"id":"cfe26ce8821c90bc59903d60d00214f8be65d1b9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-27 15:01:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/42/273042/1/check/dsvm-tempest-xen/3a6b0ad : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a45484cc007d628a5501d235dc6e51cdb15a6c5d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-27 15:30:42.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/42/273042/1/check/check-dsvm-tempest-vz7-exe-minimal/66a455b : FAILURE in 1h 05m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"8140b441b2f361b127084cd8079bdd296236399d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-27 15:42:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/1/ : SUCCESS in 2h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7e441803f0a83309436375427069103207e3461b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-27 16:17:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/42/273042/1/check/gate-nova-pep8/23af562/ : SUCCESS in 8m 13s\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/1/check/gate-nova-docs/88d500c//doc/build/html/ : SUCCESS in 11m 22s\n- gate-nova-python27 http://logs.openstack.org/42/273042/1/check/gate-nova-python27/34e4fa5/ : SUCCESS in 15m 03s\n- gate-nova-python34 http://logs.openstack.org/42/273042/1/check/gate-nova-python34/0781423/ : SUCCESS in 27m 02s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-full/e11992c/ : SUCCESS in 1h 08m 28s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-postgres-full/99cfcc6/ : SUCCESS in 1h 07m 27s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-neutron-full/0dd6f25/ : SUCCESS in 1h 05m 51s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/1/check/gate-grenade-dsvm/181f73c/ : SUCCESS in 48m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-large-ops/36de204/ : SUCCESS in 33m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-neutron-large-ops/34e02ea/ : SUCCESS in 30m 29s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/1/check/gate-nova-tox-functional/ad0b9c6/ : SUCCESS in 15m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/1/check/gate-grenade-dsvm-multinode/7db7364/ : SUCCESS in 55m 24s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6e2849c/ : FAILURE in 36m 33s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-nova-v20-api/a27b7c7/ : SUCCESS in 38m 27s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-nova-v20-api-legacy/54f245a/ : SUCCESS in 51m 21s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-cells/b9eee93/ : SUCCESS in 57m 46s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-full-ceph/7246c5d/ : SUCCESS in 45m 41s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/c7de741/ : SUCCESS in 46m 06s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/42/273042/1/check/gate-tempest-dsvm-multinode-full/4f02ad1/ : SUCCESS in 1h 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"657deb0d3b516135b1a63d4b1e4e99c531a1d872","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-27 19:50:27.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/42/273042/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"ff40a423d43e08f42ff4402d34a4d0d4d615ad77","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-27 20:32:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/273042/1/Nova-ML2-Sriov : SUCCESS in 36m 00s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"6d37b570847e3130f47bbe75b1c904fd56661b89","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-28 15:04:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/273042/1/1800 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"34112a4e3e6f0e0b78520e97cb34387e53d07e47","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-01-28 16:05:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"56a9f3c65e48c4015c86d6dd143b6db0d85c9051","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-28 16:06:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/2/ : FAILURE in 2s","accounts_in_message":[],"_revision_number":2},{"id":"17c3d1c97a174b194eaf356bce4947dac89999e8","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-28 16:13:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a637ada47b7785e5680e4e7b33bd91814c00e3eb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-28 16:18:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nNeeds a test and we probably want to handle the non-pool method that\u0027s also used in the network manager.","accounts_in_message":[],"_revision_number":2},{"id":"fedaa4b677399d4bf24318ad7adbf08065459a3e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-28 16:23:55.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSo just a -1 for a simple unit test.","accounts_in_message":[],"_revision_number":2},{"id":"144b2c6ad3e2c2e0dc7afde0d3d1148e3c62ba75","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-01-28 17:03:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da4d1863fa06d913eec5cc65d6c8a97ecbe83397","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-28 17:04:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"360323bc9bb315e79ef513f36af69f57fa225cee","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-28 17:04:50.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/3/ : FAILURE in 2s","accounts_in_message":[],"_revision_number":3},{"id":"6a68a3d03742974eb6d6c205d8d1276df7ddcb6a","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2016-01-28 17:23:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"876c607284063c6e840650e120986933cfd68b33","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-28 18:03:26.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/42/273042/3/check/check-ibm-tempest-dsvm-full/b8ccbb7/ : SUCCESS in 58m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6237b8a90022b799657300166d90766daadf151e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-28 18:12:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/42/273042/3/check/dsvm-tempest-xen/2fa7728 : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":3},{"id":"6b2943156fc476528dd2c142055eff21679249d3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-28 18:22:34.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/42/273042/3/4606/index.html : SUCCESS in 01 h 14 m 56 s","accounts_in_message":[],"_revision_number":3},{"id":"e9a24df8af7d1daa270e3c072e73cb709196ceee","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-01-28 18:22:54.000000000","message":"Patch Set 2:\n\nI proposed this back in Juno [1] but was told we should wait until the functionality is available in Neutron.\n\n[1] https://review.openstack.org/#/c/102688/","accounts_in_message":[],"_revision_number":2},{"id":"96eea44f1e7ab4156a4b6bf87564815031a3c298","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-28 19:04:29.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/273042/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"075df5979732ff91ffb7f11f8cf73553f0630563","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-28 19:13:17.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/273042/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5a6f1e5ede51c28fe4002e91ece2907b5359a525","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-28 19:25:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/42/273042/3/check/gate-nova-pep8/decddba/ : SUCCESS in 7m 31s\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/3/check/gate-nova-docs/bbfb6c1//doc/build/html/ : SUCCESS in 12m 22s\n- gate-nova-python27 http://logs.openstack.org/42/273042/3/check/gate-nova-python27/db58cb1/ : SUCCESS in 14m 07s\n- gate-nova-python34 http://logs.openstack.org/42/273042/3/check/gate-nova-python34/b90e915/ : SUCCESS in 27m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full/6c3ca9d/ : SUCCESS in 1h 00m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-postgres-full/fd92a86/ : SUCCESS in 55m 02s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-neutron-full/17ff18b/ : SUCCESS in 1h 18m 11s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/3/check/gate-grenade-dsvm/406c2ee/ : SUCCESS in 47m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-large-ops/deba735/ : SUCCESS in 44m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-neutron-large-ops/dcd545c/ : SUCCESS in 30m 34s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/3/check/gate-nova-tox-functional/5f01e63/ : SUCCESS in 15m 59s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/3/check/gate-grenade-dsvm-multinode/f8449e5/ : SUCCESS in 1h 05m 10s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2d92842/ : SUCCESS in 40m 45s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-nova-v20-api/9012bdb/ : SUCCESS in 36m 44s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-nova-v20-api-legacy/20d0d9a/ : SUCCESS in 53m 33s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-cells/2bb3f20/ : SUCCESS in 37m 31s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full-ceph/1700297/ : SUCCESS in 46m 52s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/ae226ce/ : SUCCESS in 43m 25s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-multinode-full/d604bea/ : FAILURE in 1h 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94df541e0eae98e1e1c7ac76068e9f7a3418f5b0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-28 20:04:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/273042/3/Nova-ML2-Sriov : FAILURE in 57m 21s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"944036c6db31b5ae38a78e67ab7b0747c49c1d78","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-01-28 20:10:33.000000000","message":"Patch Set 3:\n\nI just had a chat with spandhe to ask about how Neutron does fixed ip allocation and learned it uses ip ranges where it collapses the available range as it gives out ips and doesn\u0027t make the freed ips available again until all of the range is exhausted. So it does provide a \"least recently used\" functionality in that way","accounts_in_message":[],"_revision_number":3},{"id":"78f0892ad6ca8cfb903800dc727845ae3d1d73f1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-28 20:33:52.000000000","message":"Patch Set 3:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":3},{"id":"b60e999f8577d6f5d1fa2d00e1daa4a25543d7bc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-28 20:36:26.000000000","message":"Patch Set 2:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":2},{"id":"ca9eb6ef00cd8f9ffb324620df567a9ca9cfbc90","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-28 22:48:40.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/2/ : FAILURE in 2h 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"20a7dd36e715632a541ae5a6f66c1ebd5d682753","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-28 23:35:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/3/ : SUCCESS in 3h 00m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4cd574939a7dfa43434bf401c1d1bcc0657412e2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-29 00:21:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/42/273042/3/check/check-dsvm-tempest-vz7-exe-minimal/e41f04b : SUCCESS in 1h 13m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"cbacf24a9a932e48d60d361e79da9428b667ec16","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-29 01:05:32.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/273042/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9a41c4591479da56a6552e98ca31e6255fe9b213","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-29 02:39:34.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/273042/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b83046ecf22373a6ee707b32840002dde942f3ef","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-29 04:41:03.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/42/273042/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"81f125b61cb3075ebdde1d739ab0540487e40f35","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-01-29 08:58:54.000000000","message":"Patch Set 3:\n\n@mel doh, if only we\u0027d fixed it back then, so much less pain. \n\nIt\u0027s kind of amazing your spec has the same code line change.\n\nNow that nova-network isn\u0027t frozen, I think we really should move forward here. But we should circle with john because he -1ed it the last time.","accounts_in_message":[],"_revision_number":3},{"id":"3f4ff3cb3a2c0bcfe6e2ffea8e8251c735f114d2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-29 12:04:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"056a3c692470decf40ec590320857fea8e7eb156","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-29 12:04:51.000000000","message":"Patch Set 3:\n\nBTW, I read the comments re: Mel\u0027s spec, and that sucks, but it looks like it was blocked b/c nova-network was frozen at the time. It\u0027s not frozen now and this is a bug fix for the gate, which we\u0027ll probably also backport.","accounts_in_message":[],"_revision_number":3},{"id":"3aadce31c925a540628c7acacf84082415f12cdf","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-01-29 18:30:41.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good to me, I\u0027ve been wanting this change for a long time. Nice job on the test too. +1 as I\u0027d like johnthetubaguy to take a look","accounts_in_message":[],"_revision_number":3},{"id":"65048f1ac9c381f8e59c08cce8e7853e9f4c2abf","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-01-29 20:56:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdcdf6e65ff11c037d4f994293f88ea2f1dde276","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-01-29 21:14:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e5ef1e7f2c163e8942e577d5e6ad4ea41a622d3a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-30 06:30:17.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/42/273042/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7bc2207c18b741d0b821c61328af6f75/index.html : SUCCESS in 14m 42s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/42/273042/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/12b263e0aefc4a928109904808355d51/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/42/273042/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e93f4a024c9a453884e67ae7be3f09c7/index.html : SUCCESS in 12m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/42/273042/3/check/real-db-upgrade_nova_percona_user_001:th-percona/80ccf4eebc7e4ef3b699a2fdd3650c7c/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/42/273042/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/11b0fe7abbe04dbd987687bd110ec750/index.html : SUCCESS in 12m 27s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/42/273042/3/check/real-db-upgrade_nova_percona_user_002:th-percona/16b2a32af97b4624ab61c5390600b2ac/index.html : SUCCESS in 14m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"16da1948f85c7aece4cbc16f334bbc4cfc4516b6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-01-31 16:22:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/273042/3/1868 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"de859388f7d980050da83f5c90f96987242d7c70","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-01-31 23:17:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2f450fca64536fac18b468122aa0a03b3fba627","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2016-02-01 08:19:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f96130c096265cfefa53aa021abbd2e3128a20e3","author":{"_account_id":19912,"name":"Xuanzhou Perry Dong","email":"oss_xzdong@163.com","username":"oss_xzdong"},"date":"2016-02-01 09:29:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ded9f5250df52edee15f005f53a8b6a2a05ca1ed","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-02-01 11:20:58.000000000","message":"Patch Set 3: Code-Review+1\n\nThat looks all good, since Neutron does have the same behaviour, but I wonder if the added sorting could be having some performance impact that would be bad if the list of fixed ips could be huge","accounts_in_message":[],"_revision_number":3},{"id":"04aa2d0b6523c1fd88d1d757df52b58dd58b955c","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-02-01 11:36:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03a513352a4a35693533d64354271c669fdad157","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-02-01 12:00:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"982c3b4f153e1843145b3da1d722c80b22ac11b6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-02-01 13:50:59.000000000","message":"Patch Set 3: Code-Review+2\n\nSo, after thinking about that and reviewing all the calls, I find it\u0027s OK to merge it exceptionnally because the related bug is really important and I\u0027d like to see how we could find the main problem.\n\nAlso, I think that if some operators could be sad about the performance, we could probably having a following-up patch for adding an index.\n\nThat said, leaving other core(s) to +W it to make sure my opinion is agreed.","accounts_in_message":[],"_revision_number":3},{"id":"ea75f066e9d405a3c6324201878a42c35c170f0d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-02-01 14:13:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nSo we know there is risk here, but there is also risk in not doing anything to stop this issue.\n\nLets just be sure to double check that.","accounts_in_message":[],"_revision_number":3},{"id":"019cd3c621adcd0827969586184e3bd384f08614","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-01 14:13:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"09f1ef39ef5c834240bc2ab08ae27c6a7e4ec983","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-01 15:19:43.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/3/gate/gate-nova-docs/cc0b47a//doc/build/html/ : SUCCESS in 11m 59s\n- gate-nova-pep8 http://logs.openstack.org/42/273042/3/gate/gate-nova-pep8/17836af/ : SUCCESS in 7m 52s\n- gate-nova-python27 http://logs.openstack.org/42/273042/3/gate/gate-nova-python27/ec0c72f/ : SUCCESS in 17m 03s\n- gate-nova-python34 http://logs.openstack.org/42/273042/3/gate/gate-nova-python34/40e9b2c/ : SUCCESS in 27m 55s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-full/fee3f0a/ : SUCCESS in 47m 20s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-postgres-full/312fc4f/ : SUCCESS in 1h 01m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-neutron-full/9469dc0/ : SUCCESS in 59m 59s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/3/gate/gate-grenade-dsvm/7ce5fcd/ : SUCCESS in 43m 56s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-large-ops/fafa438/ : SUCCESS in 33m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-neutron-large-ops/d81bc86/ : SUCCESS in 30m 15s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/3/gate/gate-nova-tox-functional/b3de53b/ : SUCCESS in 16m 15s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/3/gate/gate-grenade-dsvm-multinode/a88469e/ : FAILURE in 46m 27s (non-voting)\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-full-ceph/e9e6343/ : FAILURE in 48m 55s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/3/gate/gate-tempest-dsvm-cells/28cb672/ : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":3},{"id":"2abc5b2599f38931f3f074d9d9a0343b14c1d266","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-02-01 15:29:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"531ee7da031781696014e6645570974fcdcb8923","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-01 16:26:03.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/42/273042/3/check/check-ibm-tempest-dsvm-full/65a3809/ : SUCCESS in 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a89e81f2182add868fe1f16f8a7c431046b5f8f3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-01 16:53:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/3/ : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":3},{"id":"4923e2d1edeb5151d6970e53ef6f1a79f0e345c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-01 17:10:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/42/273042/3/check/gate-nova-pep8/243f7c8/ : SUCCESS in 10m 11s\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/3/check/gate-nova-docs/07617fe//doc/build/html/ : SUCCESS in 18m 26s\n- gate-nova-python27 http://logs.openstack.org/42/273042/3/check/gate-nova-python27/632435a/ : SUCCESS in 14m 49s\n- gate-nova-python34 http://logs.openstack.org/42/273042/3/check/gate-nova-python34/a6ea1f9/ : SUCCESS in 28m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full/e9c796f/ : SUCCESS in 57m 04s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-postgres-full/94a8390/ : SUCCESS in 50m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-neutron-full/525003b/ : SUCCESS in 58m 14s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/3/check/gate-grenade-dsvm/42bade2/ : SUCCESS in 45m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-large-ops/2c6cd70/ : SUCCESS in 32m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-neutron-large-ops/36f28eb/ : SUCCESS in 39m 20s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/3/check/gate-nova-tox-functional/cf0f630/ : SUCCESS in 14m 25s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/3/check/gate-grenade-dsvm-multinode/10e93a2/ : SUCCESS in 58m 34s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a9fde86/ : FAILURE in 37m 42s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-nova-v20-api/e5473db/ : SUCCESS in 40m 55s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-nova-v20-api-legacy/2118730/ : FAILURE in 48m 39s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-cells/335310a/ : SUCCESS in 46m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full-ceph/9e19e31/ : SUCCESS in 57m 41s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/9c72274/ : SUCCESS in 51m 06s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-multinode-full/ff2a78e/ : SUCCESS in 1h 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dcf2698291b3fe9937c7ca3f82f36a535294e68d","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-02-01 17:34:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5d690b378bc29771dfa111d270dbebcbd19eba7d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-01 17:55:19.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/42/273042/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"09504455f0da66a7069fd48dd20772ee8b78346f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-01 18:31:09.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/42/273042/3/check/check-ibm-tempest-dsvm-full/4909d9e/ : SUCCESS in 56m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66233553eb5c5936ab23b70a92d64af3143ad258","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-01 18:56:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/42/273042/3/check/gate-nova-pep8/397d154/ : SUCCESS in 7m 39s\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/3/check/gate-nova-docs/89ea440//doc/build/html/ : SUCCESS in 12m 23s\n- gate-nova-python27 http://logs.openstack.org/42/273042/3/check/gate-nova-python27/651d897/ : SUCCESS in 15m 16s\n- gate-nova-python34 http://logs.openstack.org/42/273042/3/check/gate-nova-python34/c98c1fe/ : SUCCESS in 27m 40s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full/5864f86/ : FAILURE in 46m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-postgres-full/1df5199/ : SUCCESS in 53m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-neutron-full/0d4a6db/ : SUCCESS in 58m 08s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/3/check/gate-grenade-dsvm/099a218/ : SUCCESS in 45m 12s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-large-ops/6cbcab7/ : SUCCESS in 31m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-neutron-large-ops/5b3155a/ : SUCCESS in 35m 20s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/3/check/gate-nova-tox-functional/d9cf4cd/ : SUCCESS in 16m 09s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/3/check/gate-grenade-dsvm-multinode/75afd24/ : SUCCESS in 59m 09s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/685a8b7/ : FAILURE in 30m 33s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-nova-v20-api/95f2e17/ : SUCCESS in 37m 29s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-nova-v20-api-legacy/06afcac/ : FAILURE in 37m 55s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-cells/b9e6312/ : SUCCESS in 52m 34s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full-ceph/a55633c/ : FAILURE in 50m 56s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/29f897d/ : FAILURE in 47m 21s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/42/273042/3/check/gate-tempest-dsvm-multinode-full/9aa3480/ : FAILURE in 1h 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"deda2097fbe92c738c17b476e38fc5b304ef7fa3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-01 19:24:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/3/ : SUCCESS in 1h 49m 34s","accounts_in_message":[],"_revision_number":3},{"id":"5c22d0e0ad728bcf29598df88d206dacea66ffa7","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-02-02 01:01:56.000000000","message":"Patch Set 4: Published edit on patch set 3","accounts_in_message":[],"_revision_number":4},{"id":"bf56a70fa47932900880268621a74fff2a86f407","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 01:13:49.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/42/273042/4/check/check-dsvm-tempest-vz7-exe-minimal/9dee9bc : FAILURE in 10m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"9779cb02f454cc5d944c00d7002f200731e975b7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-02 01:24:16.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/273042/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"22f18b5484af7e6dad5785a2158b77d54720f0b9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-02 01:36:42.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/42/273042/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3a9c5b3de35044c981ed973f8361b0dd/index.html : SUCCESS in 12m 02s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/42/273042/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/16be7020317d464c83ed74d6cbe25988/index.html : SUCCESS in 19m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/42/273042/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b782534bce7e4f9a9824964390b56ac7/index.html : SUCCESS in 12m 18s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/42/273042/4/check/real-db-upgrade_nova_percona_user_001:th-percona/c0a599e63a844f4cb35669d9293013b4/index.html : SUCCESS in 15m 53s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/42/273042/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/9eefdc66f4d14ca28f342a259e7a6c9b/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/42/273042/4/check/real-db-upgrade_nova_percona_user_002:th-percona/17b3f784f2874ed09cd48d8a4cc3eff1/index.html : SUCCESS in 14m 40s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"bbbd348768c7cf1f761e48490ee1a9cfb55aec5b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-02 01:45:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/273042/4/Nova-ML2-Sriov : SUCCESS in 42m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"7418f25c60a35b656427aa6253ad74d94c0ed3a5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-02 01:59:08.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/42/273042/4/check/check-ibm-tempest-dsvm-full/11b96ba/ : SUCCESS in 56m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1db513a22f468d5a675c141be4ef8db496ca80d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 02:04:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/42/273042/4/check/gate-nova-pep8/f4cf9e0/ : SUCCESS in 8m 35s\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/4/check/gate-nova-docs/7cbfdb2//doc/build/html/ : SUCCESS in 12m 10s\n- gate-nova-python27 http://logs.openstack.org/42/273042/4/check/gate-nova-python27/ff0c85f/ : SUCCESS in 15m 12s\n- gate-nova-python34 http://logs.openstack.org/42/273042/4/check/gate-nova-python34/497a5c1/ : SUCCESS in 29m 04s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-full/0e6a5ed/ : SUCCESS in 43m 07s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-postgres-full/a465107/ : SUCCESS in 46m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-neutron-full/3c5860c/ : SUCCESS in 53m 00s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/4/check/gate-grenade-dsvm/b15a224/ : SUCCESS in 43m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-large-ops/79c51da/ : SUCCESS in 35m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-neutron-large-ops/d82d773/ : SUCCESS in 31m 57s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/4/check/gate-nova-tox-functional/4c8849a/ : SUCCESS in 17m 09s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/4/check/gate-grenade-dsvm-multinode/8594a37/ : FAILURE in 48m 42s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/48a9bd8/ : SUCCESS in 38m 16s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-nova-v20-api/a986d0c/ : SUCCESS in 38m 27s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-nova-v20-api-legacy/067bc84/ : SUCCESS in 38m 14s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-cells/e2efd19/ : SUCCESS in 44m 07s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-full-ceph/b90d674/ : SUCCESS in 41m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/278211a/ : SUCCESS in 52m 10s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/42/273042/4/check/gate-tempest-dsvm-multinode-full/4bdcac6/ : SUCCESS in 55m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a018a50b7c303beb500d4c318d0ba6edf9fe0231","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-02 02:12:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/42/273042/4/check/dsvm-tempest-xen/8e3e262 : FAILURE in 1h 09m 32s","accounts_in_message":[],"_revision_number":4},{"id":"17c19b4f1a2d42ab002878f5e9d682cad024036e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-02 02:23:53.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/42/273042/4/4862/index.html : SUCCESS in 01 h 18 m 52 s","accounts_in_message":[],"_revision_number":4},{"id":"0580ba008a024939476cdca8057c4dd90ebacc96","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-02 02:40:17.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-42-273042-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"fce45e0f0e29efa729b7f225ecfeed789fe5c48d","author":{"_account_id":11189,"name":"yuntongjin","email":"yuntong.jin@intel.com","username":"yuntong"},"date":"2016-02-02 03:26:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3ee4579db4ac0300fec6085c8bb480897f55efae","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-02 04:32:02.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/273042/4/ : FAILURE in 3h 29m 21s","accounts_in_message":[],"_revision_number":4},{"id":"c6379990d6bfd8729df1ac4c8110ef2b5b8f787b","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2016-02-02 07:28:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8d101a734af833cc29cfd1e1cfab3dad6bdccb43","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-02-02 09:11:17.000000000","message":"Patch Set 4: Code-Review+1\n\nJust a question before +2ing again, since it modifies the IP allocation, it means this is a behavioural change that breaks a Tempest test.\n\nI\u0027m fine with that, but we should probably be providing a reno change, nope ?","accounts_in_message":[],"_revision_number":4},{"id":"0dea9b8767dbb1af846a61f0922fb6e79f85d0de","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2016-02-02 10:10:11.000000000","message":"Patch Set 4:\n\nno, the Tempest test was a bad test.","accounts_in_message":[],"_revision_number":4},{"id":"f059d533a7d91fde54f74b73bef7c4967bed131a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-02-02 10:12:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nre-approve after gate issues","accounts_in_message":[],"_revision_number":4},{"id":"baa666d9449acada8011ec34cc66811d6edecef8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 10:15:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"3394e9712170b2160e5375a0ca1116a6e7c4ce74","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-02-02 10:20:06.000000000","message":"Patch Set 4: Code-Review+2\n\nYeah, the corresponding Tempest test seems odd.\nThe test just covers the specific situation ip addresses are allocated on the same network address on different servers.","accounts_in_message":[],"_revision_number":4},{"id":"64ac9800a95a1eef557a30a5b1f83bf7daba3f26","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-02-02 11:18:03.000000000","message":"Patch Set 4: Code-Review+2\n\nThen good, thanks Sean.","accounts_in_message":[],"_revision_number":4},{"id":"91c545f6b3cea468882c5fea5ebf389910dd3f5a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 11:27:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/42/273042/4/gate/gate-nova-docs/5a20e66//doc/build/html/ : SUCCESS in 11m 58s\n- gate-nova-pep8 http://logs.openstack.org/42/273042/4/gate/gate-nova-pep8/3780cec/ : SUCCESS in 7m 33s\n- gate-nova-python27 http://logs.openstack.org/42/273042/4/gate/gate-nova-python27/6c338de/ : SUCCESS in 14m 20s\n- gate-nova-python34 http://logs.openstack.org/42/273042/4/gate/gate-nova-python34/ddb0011/ : SUCCESS in 29m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-full/727c092/ : SUCCESS in 45m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-postgres-full/2263e77/ : SUCCESS in 45m 03s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-neutron-full/0fc33af/ : SUCCESS in 1h 03m 24s\n- gate-grenade-dsvm http://logs.openstack.org/42/273042/4/gate/gate-grenade-dsvm/93da21a/ : SUCCESS in 47m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-large-ops/395388b/ : SUCCESS in 31m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-neutron-large-ops/18efbe4/ : SUCCESS in 32m 46s\n- gate-nova-tox-functional http://logs.openstack.org/42/273042/4/gate/gate-nova-tox-functional/ed896c1/ : SUCCESS in 15m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/42/273042/4/gate/gate-grenade-dsvm-multinode/8d508f9/ : SUCCESS in 49m 10s (non-voting)\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-full-ceph/171be27/ : SUCCESS in 48m 33s\n- gate-tempest-dsvm-cells http://logs.openstack.org/42/273042/4/gate/gate-tempest-dsvm-cells/608986b/ : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":4},{"id":"1f2644d1593251fb7b5692c0d3077f5caf6a19c2","date":"2016-02-02 11:27:09.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"32b4ce0e92b333413163f31f2962d109941ee362","revisions":{"4613fbad5b90d02d326600a66bbe1839a3143f68":{"kind":"REWORK","_number":1,"created":"2016-01-27 13:37:51.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/42/273042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/42/273042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/42/273042/1"}}},"commit":{"parents":[{"commit":"fe74c307c3a0696c15f8201962d7987c4d4877d5","subject":"fast exit dhcpbridge on \u0027old\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe74c307c3a0696c15f8201962d7987c4d4877d5"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-27 13:02:35.000000000","tz":0},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-27 13:02:35.000000000","tz":0},"subject":"WIP: fixed ip order_by updated_at","message":"WIP: fixed ip order_by updated_at\n\nChange-Id: I3c83bd68a0e2bbbcdd6d955722dbc9f9fc528113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4613fbad5b90d02d326600a66bbe1839a3143f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4613fbad5b90d02d326600a66bbe1839a3143f68"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fe74c307c3a0696c15f8201962d7987c4d4877d5","is_merged_in_target_branch":false,"change_id":"I1539fb9eb8570a9efbaa95a1272ac9ac571c365a","change_number":273023,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9695ae01623c840a1a7659bd406ca339c2a0b9bb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2016-01-28 16:05:52.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/42/273042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/42/273042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/42/273042/2"}}},"commit":{"parents":[{"commit":"34e6c0fb15a9954247b0b7d8170a62bd37e95df1","subject":"fast exit dhcpbridge on \u0027old\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/34e6c0fb15a9954247b0b7d8170a62bd37e95df1"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-27 13:02:35.000000000","tz":0},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-28 16:04:38.000000000","tz":0},"subject":"Spread allocations of fixed ips","message":"Spread allocations of fixed ips\n\nThe default fixed_ip allocation always does .first() out of the fixed\nip pool. This leads to a pathological situation that in a rapidly\nchanging environment we\u0027re constantly recycling the same small number\nof ips out of the pool. Given other racey behavior in dnsmasq around\nprocessing DHCP Release requests, we can start to get DHCP fails when\na DHCP release is missed, we immediately recycle the IP and give it to\na new guest.\n\nWe can mitigate this by sorting the results by last_updated. This\nmeans we\u0027ll favor fixed ips that were last given out much further in\nthe past.\n\nRelated-Bug: #1532809\n\nChange-Id: I3c83bd68a0e2bbbcdd6d955722dbc9f9fc528113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9695ae01623c840a1a7659bd406ca339c2a0b9bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9695ae01623c840a1a7659bd406ca339c2a0b9bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"34e6c0fb15a9954247b0b7d8170a62bd37e95df1","is_merged_in_target_branch":true,"change_id":"I1539fb9eb8570a9efbaa95a1272ac9ac571c365a","change_number":273023,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"844bcb2bb8ae9075a200315ae3a1310d50e52763":{"kind":"REWORK","_number":3,"created":"2016-01-28 17:03:56.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/42/273042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/42/273042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/42/273042/3"}}},"commit":{"parents":[{"commit":"34e6c0fb15a9954247b0b7d8170a62bd37e95df1","subject":"fast exit dhcpbridge on \u0027old\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/34e6c0fb15a9954247b0b7d8170a62bd37e95df1"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-27 13:02:35.000000000","tz":0},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-28 17:03:03.000000000","tz":0},"subject":"Spread allocations of fixed ips","message":"Spread allocations of fixed ips\n\nThe default fixed_ip allocation always does .first() out of the fixed\nip pool. This leads to a pathological situation that in a rapidly\nchanging environment we\u0027re constantly recycling the same small number\nof ips out of the pool. Given other racey behavior in dnsmasq around\nprocessing DHCP Release requests, we can start to get DHCP fails when\na DHCP release is missed, we immediately recycle the IP and give it to\na new guest.\n\nWe can mitigate this by sorting the results by updated_at. This\nmeans we\u0027ll favor least recently used fixed ips.\n\nRelated-Bug: #1532809\n\nChange-Id: I3c83bd68a0e2bbbcdd6d955722dbc9f9fc528113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/844bcb2bb8ae9075a200315ae3a1310d50e52763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/844bcb2bb8ae9075a200315ae3a1310d50e52763"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"34e6c0fb15a9954247b0b7d8170a62bd37e95df1","is_merged_in_target_branch":true,"change_id":"I1539fb9eb8570a9efbaa95a1272ac9ac571c365a","change_number":273023,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"32b4ce0e92b333413163f31f2962d109941ee362":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2016-02-02 01:01:56.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/42/273042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/42/273042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/42/273042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/42/273042/4"}}},"commit":{"parents":[{"commit":"34e6c0fb15a9954247b0b7d8170a62bd37e95df1","subject":"fast exit dhcpbridge on \u0027old\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/34e6c0fb15a9954247b0b7d8170a62bd37e95df1"}]}],"author":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-01-27 13:02:35.000000000","tz":0},"committer":{"name":"Sean Dague","email":"sean@dague.net","date":"2016-02-02 01:01:50.000000000","tz":0},"subject":"Spread allocations of fixed ips","message":"Spread allocations of fixed ips\n\nThe default fixed_ip allocation always does .first() out of the fixed\nip pool. This leads to a pathological situation that in a rapidly\nchanging environment we\u0027re constantly recycling the same small number\nof ips out of the pool. Given other racey behavior in dnsmasq around\nprocessing DHCP Release requests, we can start to get DHCP fails when\na DHCP release is missed, we immediately recycle the IP and give it to\na new guest.\n\nWe can mitigate this by sorting the results by updated_at. This\nmeans we\u0027ll favor least recently used fixed ips.\n\nDepends-On: I27f73c1edf12218818c4d279efbd9fef5cdef672\n\nRelated-Bug: #1532809\n\nChange-Id: I3c83bd68a0e2bbbcdd6d955722dbc9f9fc528113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32b4ce0e92b333413163f31f2962d109941ee362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32b4ce0e92b333413163f31f2962d109941ee362"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"34e6c0fb15a9954247b0b7d8170a62bd37e95df1","is_merged_in_target_branch":true,"change_id":"I1539fb9eb8570a9efbaa95a1272ac9ac571c365a","change_number":273023,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
