)]}'
{"id":"openstack%2Fnova~274318","triplet_id":"openstack%2Fnova~master~I822e31bf8a739588991fcca34976d55b71b8a227","project":"openstack/nova","branch":"master","topic":"bug/1522496","hashtags":[],"change_id":"I822e31bf8a739588991fcca34976d55b71b8a227","subject":"Add instance host to evacuate terminate connector","status":"ABANDONED","created":"2016-01-30 05:15:24.000000000","updated":"2016-05-10 22:45:03.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0249694efd9626056f4b274ef38659dac0695dc1","_number":274318,"virtual_id_number":274318,"owner":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"date":"2016-03-08 12:46:29.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-07 19:47:35.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-03-07 19:59:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2016-03-08 06:13:44.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},{"value":1,"date":"2016-03-07 20:38:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2016-03-13 18:41:20.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-07 20:01:42.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-03-07 22:52:14.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-07 22:19:27.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"date":"2016-03-08 09:36:22.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"date":"2016-05-10 21:08:22.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":1,"date":"2016-03-08 16:22:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},{"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-04 17:40:14.000000000","updated_by":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"reviewer":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"state":"REVIEWER"},{"updated":"2016-02-08 13:03:36.000000000","updated_by":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"reviewer":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"state":"REVIEWER"},{"updated":"2016-02-10 17:33:15.000000000","updated_by":{"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"},"reviewer":{"_account_id":11903,"name":"Anthony Lee","email":"anthony.mic.lee@hpe.com","username":"leeantho"},"state":"REVIEWER"},{"updated":"2016-02-10 17:35:48.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2016-03-02 00:30:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-02 18:59:41.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-07 10:21:09.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-03-07 19:47:35.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-03-07 19:59:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-07 20:01:42.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-07 20:38:58.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-07 22:19:27.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-07 22:52:14.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-08 06:13:44.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-08 09:36:22.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-08 12:46:29.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-08 16:22:33.000000000","updated_by":{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},"reviewer":{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},"state":"REVIEWER"},{"updated":"2016-03-13 18:41:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-10 21:08:22.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"}],"messages":[{"id":"ffdd765fdc0acb1c2e492543c461fc0d350ac611","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-01-30 05:15:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a70c7f141c1574d3dc0167821dfb9feda76a99cf","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 05:32:55.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/274318/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0b6ccd09e9df66dba8d678aa4a19db1253425040","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-30 05:49:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/1/Nova-ML2-Sriov : SUCCESS in 32m 26s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1330c107daea1e2d5ef6712309b29df1d8be179","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 06:05:18.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"44cdda2548bbb1b107eadbd2d3de4ba5d2783923","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-30 06:12:00.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/1/check/check-ibm-tempest-dsvm-full/3966749/ : SUCCESS in 55m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8ed9eff694db5d0e44fcebdc7b1a1e79670a5b9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-30 06:25:48.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/1/4723/index.html : SUCCESS in 01 h 06 m 55 s","accounts_in_message":[],"_revision_number":1},{"id":"233a049a73d0941115358f84233bc7abea8de7b2","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-30 06:30:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/1/check/dsvm-tempest-xen/f596ecb : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"19ff564b6bad2840316d3cb22bbe24fe2467d639","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-30 06:35:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/1/check/gate-nova-pep8/1380ed9/ : SUCCESS in 8m 22s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/1/check/gate-nova-docs/7aebe2d//doc/build/html/ : SUCCESS in 14m 05s\n- gate-nova-python27 http://logs.openstack.org/18/274318/1/check/gate-nova-python27/e8a75b9/ : SUCCESS in 15m 36s\n- gate-nova-python34 http://logs.openstack.org/18/274318/1/check/gate-nova-python34/d105137/ : SUCCESS in 26m 24s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-full/947013d/ : SUCCESS in 52m 32s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-postgres-full/4dee24b/ : SUCCESS in 42m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-neutron-full/41e996e/ : SUCCESS in 56m 12s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/1/check/gate-grenade-dsvm/6a983ba/ : SUCCESS in 43m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-large-ops/e8ebfeb/ : SUCCESS in 31m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-neutron-large-ops/1e0afa5/ : SUCCESS in 31m 00s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/1/check/gate-nova-tox-functional/09d1e4e/ : SUCCESS in 14m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/1/check/gate-grenade-dsvm-multinode/ac62d79/ : FAILURE in 45m 15s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6fc52cc/ : FAILURE in 31m 18s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-nova-v20-api/aa6f416/ : SUCCESS in 37m 21s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-nova-v20-api-legacy/e017202/ : SUCCESS in 36m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-cells/23b34c2/ : SUCCESS in 40m 18s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-full-ceph/d968abd/ : SUCCESS in 47m 11s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/9178bdc/ : SUCCESS in 47m 07s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/1/check/gate-tempest-dsvm-multinode-full/9c884b1/ : FAILURE in 1h 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aae81d6eff99f1ed67ad97558dee34a7b17dfb61","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 06:38:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/1/check/check-dsvm-tempest-vz7-exe-minimal/0bd9f2b : SUCCESS in 1h 21m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"e8f238c21021267b84dc130b87e52d98b2925158","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-01-30 06:43:27.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-274318-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"01491b85ef2204c846a140c95698a2d51fbab64a","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-01-30 06:49:30.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7a4ecc6ad97311c8b07988c6db11cf98f14f5d9c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 06:58:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/1/ : SUCCESS in 1h 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"deb74a20629e1a0adb218f06bb95ca7e1b24ad50","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-01-30 07:41:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6a0125a8d455d36781d9e82fdd3befa055a3609c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 07:59:33.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/274318/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"67b5688dffb348472dd56b4e9c822a94fb0a7749","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-30 08:16:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/2/Nova-ML2-Sriov : SUCCESS in 33m 12s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"04feda530c92363874c450a35599fdc5fce11b3e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 08:31:23.000000000","message":"Patch Set 2:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/2 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"7e12bb1da6d417f2ea32a9a3a3438efa6d73ac28","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-30 08:39:09.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/2/check/check-ibm-tempest-dsvm-full/ee1192f/ : SUCCESS in 56m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46aee7de3b9bf877911e87034db2628412ca3c30","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-30 08:53:36.000000000","message":"Patch Set 2:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/2/4727/index.html : SUCCESS in 01 h 08 m 25 s","accounts_in_message":[],"_revision_number":2},{"id":"41de440500239f3c5b99fa7e3b10acf89d4be7f1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-30 08:56:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/2/check/dsvm-tempest-xen/635d663 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ee1a34498f3e1449f676c0bec7466e254b6e08a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-30 09:01:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/2/check/gate-nova-pep8/9cc673d/ : SUCCESS in 7m 22s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/2/check/gate-nova-docs/3ab0ce6//doc/build/html/ : SUCCESS in 11m 31s\n- gate-nova-python27 http://logs.openstack.org/18/274318/2/check/gate-nova-python27/50fbb1b/ : SUCCESS in 15m 35s\n- gate-nova-python34 http://logs.openstack.org/18/274318/2/check/gate-nova-python34/8e5e472/ : SUCCESS in 26m 14s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-full/9f5b2b9/ : SUCCESS in 42m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-postgres-full/7539836/ : SUCCESS in 52m 17s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-neutron-full/170af6c/ : SUCCESS in 58m 27s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/2/check/gate-grenade-dsvm/2e9a3be/ : SUCCESS in 42m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-large-ops/773874b/ : SUCCESS in 35m 10s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-neutron-large-ops/da8369a/ : SUCCESS in 28m 47s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/2/check/gate-nova-tox-functional/53e278b/ : SUCCESS in 15m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/2/check/gate-grenade-dsvm-multinode/68b0b91/ : FAILURE in 46m 22s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6dc3953/ : FAILURE in 29m 12s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-nova-v20-api/17b857b/ : SUCCESS in 35m 38s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-nova-v20-api-legacy/c1c658d/ : SUCCESS in 33m 04s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-cells/e21a7cd/ : SUCCESS in 41m 54s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-full-ceph/ab5240e/ : SUCCESS in 45m 27s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/bc80ecd/ : SUCCESS in 52m 24s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/2/check/gate-tempest-dsvm-multinode-full/6b63368/ : FAILURE in 1h 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2da6dc1de3290c0c8dad9965127117929900152b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 09:10:21.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/2/check/check-dsvm-tempest-vz7-exe-minimal/98cc14d : FAILURE in 1h 27m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"9d747ecc0414ff9a2f98dadee1323a836f242e2a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 10:42:11.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/2/ : FAILURE in 2h 59m 48s","accounts_in_message":[],"_revision_number":2},{"id":"cbc2f8ed5d220d3fcb380a2b13726a76b7f2fade","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-31 00:08:49.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2d16891e303e49d0ae8acddf7de2c947/index.html : SUCCESS in 14m 29s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/bd477bdfba3a46fcbf413e9aaf57e702/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d3c87a2399ae4983996b36120b57ecd1/index.html : SUCCESS in 12m 36s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/2/check/real-db-upgrade_nova_percona_user_001:th-percona/cfbafd0e82b347cf8bf150ffdee7f735/index.html : SUCCESS in 17m 17s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f9646b41a757471faafe06d54f7020ff/index.html : SUCCESS in 12m 26s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/2/check/real-db-upgrade_nova_percona_user_002:th-percona/ee97efb006ef4c6883e11f2d2656df03/index.html : SUCCESS in 14m 23s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"310d4edff46b47947c49364fae990d3e44518718","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-03 06:15:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/274318/2/1922 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c2ea3abc7723a17c83ab5b8d671a522923d9c09f","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-03 21:29:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3f6b74a292bdae94b7592e83f5efb2f741a5bb37","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-03 21:46:56.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/274318/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"7ec001f7ecd8d30d7001c2b76a7fb8480ecdeb5d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-03 21:55:07.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/2e44c5ed522542c0835e6602d0e53272/index.html : SUCCESS in 14m 22s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5edf9cfb647a4ccb97fd4811e3f9fe73/index.html : SUCCESS in 12m 58s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9e84ec82638a4a9d8b1c29fcf9bbcf47/index.html : SUCCESS in 15m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/3/check/real-db-upgrade_nova_percona_user_001:th-percona/b18c23d486724160b285425c8563467a/index.html : SUCCESS in 13m 08s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/e64c33584e5e4232bcdce78d456c0c19/index.html : SUCCESS in 14m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/3/check/real-db-upgrade_nova_percona_user_002:th-percona/3c1425565294489ea0cba063ffcb05b7/index.html : SUCCESS in 14m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"96bba9303854b9e8f39dfebbbdace0dbd0ed403d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-03 22:27:04.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/3/check/check-ibm-tempest-dsvm-full/d65a726/ : SUCCESS in 56m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8fee98c022b32cbc12abc3ed0f47be294035abd0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-03 22:41:02.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/3/5172/index.html : SUCCESS in 01 h 08 m 25 s","accounts_in_message":[],"_revision_number":3},{"id":"3a1df3ba001029d70121b2f81327a765ac8ce85f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-03 22:46:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/3/check/dsvm-tempest-xen/a3eb6ea : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":3},{"id":"f178484973c569f39b8d8f075e362fdd5f36d58a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-04 01:01:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/3/check/gate-nova-pep8/8982a24/ : SUCCESS in 7m 58s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/3/check/gate-nova-docs/d61b22e//doc/build/html/ : SUCCESS in 12m 41s\n- gate-nova-python27 http://logs.openstack.org/18/274318/3/check/gate-nova-python27/18d9458/ : SUCCESS in 16m 53s\n- gate-nova-python34 http://logs.openstack.org/18/274318/3/check/gate-nova-python34/75a248e/ : SUCCESS in 27m 33s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-full/3d02444/ : SUCCESS in 46m 57s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-postgres-full/0d0d13b/ : SUCCESS in 48m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-neutron-full/b6cbd4c/ : SUCCESS in 1h 02m 10s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/3/check/gate-grenade-dsvm/2579ac2/ : SUCCESS in 47m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-large-ops/4338baa/ : SUCCESS in 37m 04s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-neutron-large-ops/e7b674a/ : SUCCESS in 31m 42s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/3/check/gate-nova-tox-functional/17a4d51/ : SUCCESS in 17m 32s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/3/check/gate-grenade-dsvm-multinode/e15d839/ : SUCCESS in 57m 04s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a737efb/ : SUCCESS in 53m 33s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-nova-v20-api/982d864/ : SUCCESS in 36m 49s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-nova-v20-api-legacy/02fb75a/ : SUCCESS in 36m 20s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-cells/be3429f/ : SUCCESS in 36m 04s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-full-ceph/be08068/ : SUCCESS in 1h 20m 39s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/3a7a3d8/ : SUCCESS in 1h 20m 33s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/3/check/gate-tempest-dsvm-multinode-full/148ea9f/ : FAILURE in 1h 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b8d0884d26a0998cd1300586145b55e598edb626","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-04 01:38:20.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-274318-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"e39956931871e3935a7497d4db5f9dea8b3b27d4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 02:41:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/3/ : SUCCESS in 1h 56m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a47bf0f873e9cacd409a8246308fe353c9befa7b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-04 03:34:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/3/Nova-ML2-Sriov : SUCCESS in 41m 14s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"4868dfb01c5fe94f0a66c1a9da93f290a69cc12f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 05:40:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/3/check/check-dsvm-tempest-vz7-exe-minimal/da24a61 : SUCCESS in 1h 28m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"c2a4a4e8a065a1d553bbac8bbaf3dfdccaa5f5f7","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-06 05:54:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dacacb3c373bdd3c1d5e27edf02e254f236f4172","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-06 06:11:08.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/25a5d0e703154d1babd59f6140ee8766/index.html : SUCCESS in 11m 03s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/13a582ff02f74de08d47d55e66e585e8/index.html : SUCCESS in 15m 19s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1cf0219f709d4a0b8bf84ce9ea61355b/index.html : SUCCESS in 14m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/4/check/real-db-upgrade_nova_percona_user_001:th-percona/4dec097dd5114fea9c920a7be14f493f/index.html : SUCCESS in 15m 13s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/334b64f0a70d4752b1f8d1ff8fa4eb35/index.html : SUCCESS in 14m 15s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/4/check/real-db-upgrade_nova_percona_user_002:th-percona/91f2c1cbb8554af4aa8600c4e76f1d6b/index.html : SUCCESS in 11m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"6fe80b3e39ee947b88abde57cb614423f84e2dde","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-06 06:12:14.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/274318/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ecccdad96a900b095d03ec6515eb355a09f1c3d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-06 06:55:08.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/4/check/check-ibm-tempest-dsvm-full/ba66432/ : SUCCESS in 59m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7dcdc0ba3cad3c888dd13898e0a5d184502f5c1b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-06 06:58:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/4/check/gate-nova-pep8/51bc72a/ : SUCCESS in 2m 38s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/4/check/gate-nova-docs/779b10e//doc/build/html/ : SUCCESS in 8m 27s\n- gate-nova-python27 http://logs.openstack.org/18/274318/4/check/gate-nova-python27/532c1b4/ : FAILURE in 9m 59s\n- gate-nova-python34 http://logs.openstack.org/18/274318/4/check/gate-nova-python34/3774e93/ : FAILURE in 27m 33s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-full/4982808/ : SUCCESS in 55m 15s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-postgres-full/4a91371/ : SUCCESS in 33m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-neutron-full/1596482/ : SUCCESS in 1h 00m 36s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/4/check/gate-grenade-dsvm/47c5ecf/ : SUCCESS in 42m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-large-ops/3cd074e/ : SUCCESS in 28m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-neutron-large-ops/d2dae76/ : SUCCESS in 27m 50s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/4/check/gate-nova-tox-functional/cece442/ : SUCCESS in 10m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/4/check/gate-grenade-dsvm-multinode/f90f8ca/ : SUCCESS in 34m 29s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6959b24/ : SUCCESS in 30m 39s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-nova-v20-api/1ff100a/ : SUCCESS in 28m 46s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-nova-v20-api-legacy/5847905/ : SUCCESS in 29m 37s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-cells/fc5e943/ : SUCCESS in 31m 36s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-full-ceph/7faae30/ : SUCCESS in 41m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/332d74f/ : SUCCESS in 43m 56s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/4/check/gate-tempest-dsvm-multinode-full/b8f618c/ : SUCCESS in 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"efee87c7d5151f34b2dad1837e9d7383b7961f41","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-06 07:19:19.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/4/5452/index.html : SUCCESS in 01 h 19 m 01 s","accounts_in_message":[],"_revision_number":4},{"id":"e5caf3cdbeeb5e4c7505353843f96e738d1e16c6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-06 08:24:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/4/check/check-dsvm-tempest-vz7-exe-minimal/3ed55f3 : SUCCESS in 1h 16m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"8e491605a3fb74bb967e3a1688b84161ef68fe77","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-06 09:27:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/4/ : FAILURE in 3h 32m 33s","accounts_in_message":[],"_revision_number":4},{"id":"f21176e0f15b914384fb4246eb328ea8ff857c21","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-06 20:01:51.000000000","message":"Patch Set 4:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/4 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":4},{"id":"62426dc70381610ced1f485db147215dd3e4e7cd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-07 00:51:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/4/Nova-ML2-Sriov : FAILURE in 2h 22m 34s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"034c8832e273d7c70da076c7a6bfa2300232e236","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-07 14:15:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/4/Nova-ML2-Sriov : SUCCESS in 35m 14s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"e5816971659ae9b60397875aad6e813193c892d7","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-08 17:47:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"da31ea9d48f4dae7b346f29a9bdd176a04b6a132","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-08 18:46:43.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/5/check/check-ibm-tempest-dsvm-full/fff29e7/ : FAILURE in 58m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e9989a8d80ba18d0ed1ac93dbd95f420c5c424f3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-08 18:50:58.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/274318/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"4d0908dc745dbea8c80cbb5902ae202428ed3f8f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-08 19:07:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/5/check/dsvm-tempest-xen/5076ab4 : FAILURE in 1h 18m 45s","accounts_in_message":[],"_revision_number":5},{"id":"27175eb2ab6fe40e7cc5595c138c751e6e8d57dd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-08 19:11:18.000000000","message":"Patch Set 5: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/5/5594/index.html : FAILURE in 01 h 18 m 23 s","accounts_in_message":[],"_revision_number":5},{"id":"1338ebc72f6b4b3f04ba6a400d06cad26d5847b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-08 20:08:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/5/check/gate-nova-pep8/5ae2c45/ : SUCCESS in 2m 16s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/5/check/gate-nova-docs/2c0decc//doc/build/html/ : SUCCESS in 7m 32s\n- gate-nova-python27 http://logs.openstack.org/18/274318/5/check/gate-nova-python27/5f61780/ : SUCCESS in 9m 55s\n- gate-nova-python34 http://logs.openstack.org/18/274318/5/check/gate-nova-python34/675337d/ : SUCCESS in 28m 13s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-full/99f2920/ : SUCCESS in 48m 50s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-postgres-full/38dcf73/ : SUCCESS in 37m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-neutron-full/e564497/ : SUCCESS in 1h 17m 17s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/5/check/gate-grenade-dsvm/83dc1f8/ : SUCCESS in 42m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-large-ops/770fb54/ : SUCCESS in 25m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-neutron-large-ops/7f7d9bf/ : SUCCESS in 31m 39s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/5/check/gate-nova-tox-functional/46584c1/ : SUCCESS in 12m 16s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/5/check/gate-grenade-dsvm-multinode/6a2f6ab/ : SUCCESS in 41m 12s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ae7d40b/ : SUCCESS in 40m 22s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-nova-v20-api/6352d31/ : SUCCESS in 27m 36s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-nova-v20-api-legacy/37e947f/ : SUCCESS in 36m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-cells/8b18f3b/ : FAILURE in 29m 52s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-full-ceph/ca4253b/ : FAILURE in 40m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/779ca37/ : FAILURE in 37m 17s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-multinode-full/52f3522/ : SUCCESS in 59m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d8c43d819369ffa2ef21a00dd38de3d56afadc3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-08 20:28:51.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/29e4eb6bafd647e9945856de763364f0/index.html : SUCCESS in 15m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/64731c39fd2b416ea5215cb8e8ef9cf8/index.html : SUCCESS in 15m 29s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b976b5f899d741deabc60679c9c21c9a/index.html : SUCCESS in 12m 52s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/5/check/real-db-upgrade_nova_percona_user_001:th-percona/d60a395f03af4dc4b1cd328a6570f74f/index.html : SUCCESS in 18m 51s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/f1077d313ed54ce7a072e6f0c19398e0/index.html : SUCCESS in 12m 44s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/5/check/real-db-upgrade_nova_percona_user_002:th-percona/9bc5677fcbdf41a788d48df564e43e5e/index.html : SUCCESS in 11m 52s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"63f4948512699f4b26d8b6570930560ddd1e07ee","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-08 21:53:28.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/5/ : FAILURE in 1h 50m 49s","accounts_in_message":[],"_revision_number":5},{"id":"e66b6af26ed723b3aaa7f55db87d8148d9a6a250","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 01:47:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/5/check/check-dsvm-tempest-vz7-exe-minimal/b61a613 : SUCCESS in 1h 04m 47s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"b4c78c2bf3c84af2e1f919b6e4bdea0b9a40b21a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-09 02:44:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/5/Nova-ML2-Sriov : FAILURE in 11m 56s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"5c8b3670b416c459dc23bc2993935afc2f55a89e","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-09 05:29:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b8aa278d14f36757f5429b3b3bad3147ff89c0c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-09 06:29:52.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/5/check/check-ibm-tempest-dsvm-full/d4d28eb/ : SUCCESS in 59m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"615b85d294f6810a296865b511b856dfa073a868","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-09 06:41:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/5/check/gate-nova-pep8/d8f9764/ : SUCCESS in 2m 53s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/5/check/gate-nova-docs/03b2691//doc/build/html/ : SUCCESS in 4m 49s\n- gate-nova-python27 http://logs.openstack.org/18/274318/5/check/gate-nova-python27/09d3fa2/ : SUCCESS in 10m 23s\n- gate-nova-python34 http://logs.openstack.org/18/274318/5/check/gate-nova-python34/b4b1162/ : SUCCESS in 29m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-full/0793d93/ : SUCCESS in 35m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-postgres-full/bda8a99/ : SUCCESS in 35m 01s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-neutron-full/2fcc824/ : SUCCESS in 50m 34s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/5/check/gate-grenade-dsvm/2919108/ : SUCCESS in 36m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-large-ops/360e6eb/ : SUCCESS in 23m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-neutron-large-ops/afe0df8/ : SUCCESS in 21m 28s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/5/check/gate-nova-tox-functional/7a138d6/ : SUCCESS in 12m 24s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/5/check/gate-grenade-dsvm-multinode/371a1ca/ : SUCCESS in 45m 01s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3f692c2/ : SUCCESS in 37m 27s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-nova-v20-api/ce4782a/ : SUCCESS in 30m 32s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-nova-v20-api-legacy/fa42af4/ : SUCCESS in 27m 56s\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-cells/9e58219/ : SUCCESS in 34m 17s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-full-ceph/0cfe403/ : SUCCESS in 41m 46s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/d512904/ : SUCCESS in 41m 33s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/5/check/gate-tempest-dsvm-multinode-full/8e5b9e4/ : FAILURE in 1h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5db15fcda5bbb2b8a57b5dcadaae15f18f9bdfc3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 06:49:51.000000000","message":"Patch Set 5:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/5 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"c0cdffcc2a3be1bab414d5813af527ec1158dfd5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-09 07:21:58.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-274318-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"3ca34829b12dd1b533eb9899516a881295168ab2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 07:31:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/5/ : SUCCESS in 2h 01m 16s","accounts_in_message":[],"_revision_number":5},{"id":"e9ea196e2087e25dcec75ed53a375399721407c8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 15:37:22.000000000","message":"Patch Set 5:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/5 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"88745852129905a8615b8c21c63ddd009c5032bb","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-09 16:58:13.000000000","message":"Patch Set 5:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":5},{"id":"f307819c1cdb29ee5e86c1200763b6c7959da67f","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-09 16:58:28.000000000","message":"Patch Set 5:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":5},{"id":"58b43c184012f0b2181d25e9970e714bb245fb09","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-09 17:12:31.000000000","message":"Patch Set 5:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/5/5710/index.html : FAILURE in 10 m 36 s","accounts_in_message":[],"_revision_number":5},{"id":"99cf530be851e372e40f600dfecda916a1f639b3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 17:43:15.000000000","message":"Patch Set 5:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/5 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"5b6ec777c3dbddbdaed459ad6aaff7f92a1dbf17","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-10 17:32:06.000000000","message":"Patch Set 5:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":5},{"id":"1ade0b636c47134a6b26a380204fbf0ade6249f5","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-10 17:32:34.000000000","message":"Patch Set 5:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":5},{"id":"9e045955bbf733955da73cb799e67d7a4d36d940","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-10 17:38:11.000000000","message":"Patch Set 5:\n\nSee example Cinder patches to make nova evacuate work with pre-existing attachments using this patch:\n\n3PAR - With FibreChannel Zone Manager, FC and iSCSI:\nhttps://review.openstack.org/266098\nLeftHand - Simple use of host:\nhttps://review.openstack.org/272899","accounts_in_message":[],"_revision_number":5},{"id":"7bebc1a46c1198dc71301d3854d74bb1d79319ab","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-10 19:00:53.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/5/5897/index.html : SUCCESS in 01 h 18 m 30 s","accounts_in_message":[],"_revision_number":5},{"id":"92ee289526718a0e445ba9ce04b59e53d83f6952","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-11 04:17:20.000000000","message":"Patch Set 5:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/5 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"b52828982b1fe2b6e6878bc3494ac69407ef434f","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-11 07:39:48.000000000","message":"Patch Set 5:\n\nxen: recheck","accounts_in_message":[],"_revision_number":5},{"id":"0b4a03bc4fd004efb7473800b7201caaf0ec0d30","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-12 03:25:41.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ebe102a1ba37195a07891800db6da5833f036326","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-12 03:39:55.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/274318/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"39b56c4b9e02f0525adc97acae636cd025312dc3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-12 03:44:33.000000000","message":"Patch Set 6:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a15186b0a5c844d68fcfa871c8fa637a/index.html : SUCCESS in 12m 12s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f0dfed7c19644e588286b71ceefe8dfc/index.html : SUCCESS in 13m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9d0e7606338c426faabf5ef4278d4898/index.html : SUCCESS in 14m 30s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/6/check/real-db-upgrade_nova_percona_user_001:th-percona/439600ec040c4f45a5992cc40e2930f0/index.html : SUCCESS in 17m 36s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/777694ba6d164a2fa0e69cf49b0e01eb/index.html : SUCCESS in 13m 53s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/6/check/real-db-upgrade_nova_percona_user_002:th-percona/9e4cca4a39104b8baf723d2422dc3b68/index.html : SUCCESS in 15m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"27f64d3e4771cdbbcb791a1d5ab0d1604893bbac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-12 04:29:46.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/6/check/check-ibm-tempest-dsvm-full/b6fa7d4/ : SUCCESS in 1h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"edc29395015176eebb2bcf357027beffa898a544","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 04:31:11.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/6/check/check-dsvm-tempest-vz7-exe-minimal/dbccc6e : FAILURE in 1h 04m 30s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"f49027dc4699c07406336a73c3eb627c516525a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 04:34:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/6/check/gate-nova-pep8/8ecbaa7/ : SUCCESS in 2m 25s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/6/check/gate-nova-docs/c970aaf//doc/build/html/ : SUCCESS in 7m 50s\n- gate-nova-python27 http://logs.openstack.org/18/274318/6/check/gate-nova-python27/2d07ec4/ : SUCCESS in 10m 53s\n- gate-nova-python34 http://logs.openstack.org/18/274318/6/check/gate-nova-python34/1e51498/ : SUCCESS in 32m 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-full/1e34ab5/ : SUCCESS in 38m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-postgres-full/1d6f867/ : SUCCESS in 50m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-neutron-full/796726e/ : SUCCESS in 46m 09s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/6/check/gate-grenade-dsvm/b2bd896/ : SUCCESS in 31m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-large-ops/9638d6f/ : SUCCESS in 24m 40s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-neutron-large-ops/a4497e9/ : SUCCESS in 22m 10s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/6/check/gate-nova-tox-functional/3eb0ca4/ : SUCCESS in 13m 07s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/6/check/gate-grenade-dsvm-multinode/8454a77/ : SUCCESS in 38m 22s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/387c1a5/ : SUCCESS in 30m 30s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-cells/09c986e/ : SUCCESS in 29m 25s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-full-ceph/ce78b8b/ : SUCCESS in 36m 45s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/02d369a/ : SUCCESS in 49m 56s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/6/check/gate-tempest-dsvm-multinode-full/c7f8d4f/ : FAILURE in 57m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"453515ce4568fc4f4d13db35e2420a32132559e2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-12 04:46:44.000000000","message":"Patch Set 6:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/6/6120/index.html : SUCCESS in 01 h 16 m 38 s","accounts_in_message":[],"_revision_number":6},{"id":"65dacf2bc2c028c65c397a96c1ca9818a21c88f4","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-12 04:47:41.000000000","message":"Patch Set 7: Published edit on patch set 6","accounts_in_message":[],"_revision_number":7},{"id":"94ca3c35d100ddef0121c59ca3dcc40036e8150e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-12 05:04:39.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/77685c20abc245f0af74939136492312/index.html : SUCCESS in 13m 36s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0e18ba6c4a58480083fe4f51ec3b2083/index.html : SUCCESS in 12m 12s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dcf9ad8205a84a1384be29ed39d87844/index.html : SUCCESS in 11m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/7/check/real-db-upgrade_nova_percona_user_001:th-percona/be506e0a871c4837a216914844bca9ba/index.html : SUCCESS in 15m 47s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7c58a3d91d764fb78a17bb7ced037784/index.html : SUCCESS in 13m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/7/check/real-db-upgrade_nova_percona_user_002:th-percona/9c4bd6ddfe494cabb3b5436d7f28870e/index.html : SUCCESS in 11m 52s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"b214f6a94a6453f7ae16bf50f532f5b65cd9fdce","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-12 05:06:07.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/274318/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"45ea927aa346fffd82b65a2fa52f8f172d834bab","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 05:44:53.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/7/check/check-dsvm-tempest-vz7-exe-minimal/f82b7ff : FAILURE in 56m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"0727b58f71ee4e9a5dd7df984e4da429100797d4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-12 05:49:28.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/7/check/check-ibm-tempest-dsvm-full/ab07e84/ : SUCCESS in 1h 00m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"46adb392fdc57b4ea56320f884dfd137bb3f37c6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 06:05:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/7/check/gate-nova-pep8/f06c8ac/ : SUCCESS in 2m 40s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/7/check/gate-nova-docs/6268d01//doc/build/html/ : SUCCESS in 7m 21s\n- gate-nova-python27 http://logs.openstack.org/18/274318/7/check/gate-nova-python27/1f1f113/ : SUCCESS in 10m 25s\n- gate-nova-python34 http://logs.openstack.org/18/274318/7/check/gate-nova-python34/5910f78/ : SUCCESS in 27m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-full/ac69e37/ : SUCCESS in 40m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-postgres-full/edc9d1a/ : SUCCESS in 46m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-neutron-full/9fc94ed/ : SUCCESS in 1h 03m 04s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/7/check/gate-grenade-dsvm/b3ba5d5/ : SUCCESS in 41m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-large-ops/35ab034/ : SUCCESS in 23m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-neutron-large-ops/7eb68ad/ : SUCCESS in 24m 35s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/7/check/gate-nova-tox-functional/2a834c8/ : SUCCESS in 12m 48s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/7/check/gate-grenade-dsvm-multinode/e1cbaee/ : SUCCESS in 50m 48s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9a6a7ea/ : SUCCESS in 36m 49s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-cells/713a4cd/ : SUCCESS in 59m 20s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-full-ceph/f89f6fa/ : SUCCESS in 40m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/729f2d2/ : SUCCESS in 46m 49s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/7/check/gate-tempest-dsvm-multinode-full/aa19c64/ : FAILURE in 1h 07m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e280f445cfec3a3053452f56a3ba56eaa3409572","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-12 06:07:03.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/7/6123/index.html : SUCCESS in 01 h 13 m 33 s","accounts_in_message":[],"_revision_number":7},{"id":"f925fd6e56724d749f7bfbddf3c45b61fb7f0596","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-12 06:32:40.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-274318-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"82d2bf76ad2c546f1d0d5eedcbef264ba2397bd2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 08:17:07.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/7/ : FAILURE in 3h 28m 37s","accounts_in_message":[],"_revision_number":7},{"id":"13cb05576f22371798aca4df367efb6eecf27df6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-12 15:43:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/7/Nova-ML2-Sriov : SUCCESS in 1h 08m 10s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"e6475da518c45ff2b522053198daa40fe6a16321","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 18:11:03.000000000","message":"Patch Set 6:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/6 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":6},{"id":"80ca9ee51738e9f443db8aab228860479feffceb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-17 00:36:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/274318/7/2284 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"1490989fbc6f2fad0389a91d4be0d7f3692fde1e","author":{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},"date":"2016-02-29 09:48:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6936d8d53f70fee022407dd08ca3b07e0ad863ed","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-02-29 18:12:21.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nreply inline","accounts_in_message":[],"_revision_number":7},{"id":"29bedf33c7132cb5b08bdfd3197f58d353a444db","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-03-01 22:00:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"4acdae846259631fe41f79a9fc0bcb584b17e4a4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-01 22:17:01.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/8/ : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":8},{"id":"81c1afe6769ad6cdaf53a99fa2731136338a60fb","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-01 22:18:44.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/274318/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"8590e72e7cacf54a340f4bad036e42676a137275","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-01 22:21:54.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7f59ae2a885a493499263b5bf9467ca5/index.html : SUCCESS in 12m 52s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c4cf95df4ad24534bcd7821673594442/index.html : SUCCESS in 16m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a0e74dbb244a4ddeb57b1d1a0109fb4b/index.html : SUCCESS in 19m 36s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/8/check/real-db-upgrade_nova_percona_user_001:th-percona/ab39882ae6db4a4292199dcd70b61e67/index.html : SUCCESS in 14m 10s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/12197fd23d924b5b89c0f8390b02a75a/index.html : SUCCESS in 13m 03s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/8/check/real-db-upgrade_nova_percona_user_002:th-percona/f9ba746b0ce846d3b2ab446fcf6a3b22/index.html : SUCCESS in 13m 21s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"3897a3e036335faadf89a7ef7aaea606ca749974","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-01 23:03:06.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/8/check/check-ibm-tempest-dsvm-full/3110d23/ : SUCCESS in 1h 00m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6a1221841557504977d2323a1d39f0b188aef70e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-01 23:11:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/8/check/dsvm-tempest-xen/278f77f : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":8},{"id":"4f1a11a5796fb3e3fdd047277ab76bdef3417e16","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-01 23:33:40.000000000","message":"Patch Set 8: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/8/7964/index.html : SUCCESS in 01 h 28 m 31 s","accounts_in_message":[],"_revision_number":8},{"id":"916c7307ba64835e27eae1be14ca2551e3f8f4ed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-02 00:30:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/8/check/gate-nova-pep8/dccebed/ : SUCCESS in 2m 08s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/8/check/gate-nova-docs/d77a6f2//doc/build/html/ : SUCCESS in 7m 04s\n- gate-nova-python27 http://logs.openstack.org/18/274318/8/check/gate-nova-python27/65fb80b/ : SUCCESS in 8m 39s\n- gate-nova-python34 http://logs.openstack.org/18/274318/8/check/gate-nova-python34/0ce6d9e/ : SUCCESS in 30m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-full/084c283/ : SUCCESS in 56m 24s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-postgres-full/c191a6f/ : SUCCESS in 55m 48s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-neutron-full/1f4b896/ : SUCCESS in 1h 13m 21s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/8/check/gate-grenade-dsvm/59f1fab/ : SUCCESS in 33m 49s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/8/check/gate-nova-tox-functional/2b0e8c8/ : SUCCESS in 11m 37s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/8/check/gate-grenade-dsvm-multinode/a65e250/ : SUCCESS in 55m 47s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/935bb2c/ : SUCCESS in 37m 23s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-cells/ac28d3e/ : SUCCESS in 34m 30s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-full-ceph/cea7234/ : SUCCESS in 46m 49s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/fa3b8d9/ : SUCCESS in 38m 33s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/8/check/gate-tempest-dsvm-multinode-full/f94847c/ : SUCCESS in 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"19c1231feb2ae0fc2a0c745e8744792f0c2b6374","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-03-02 00:30:42.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-274318-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"7407645d95e46f24cf89ad2a316c26b3ff127aea","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-02 04:33:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/8/check/check-dsvm-tempest-vz7-exe-minimal/d34e458 : SUCCESS in 1h 59m 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"2c610f6d055d4bb2a7dd91c6511e0afe250f4970","author":{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},"date":"2016-03-02 10:48:23.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d4a73931ec670ce69a6376d211471a1bcc4d14b1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-03-02 18:59:41.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274318/8/Nova-ML2-Sriov : SUCCESS in 58m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"d0947bc1efd8ec2423ca53d8d46959b02809693c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 07:31:15.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"c78299171c3a234c428b6a3d6a4e906a94ba0f2a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-03-07 10:10:02.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"5ce3f4ac313ec2ccd7ee09b14e2573f837c90705","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-03-07 10:21:09.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)\n\nSo this looks like the best approach.\n\nJust got some nits around the logging here. Probably good to warn folks about what is going on here.","accounts_in_message":[],"_revision_number":9},{"id":"a82faa1a96bd222cf288c5acfbf454a5fc30eecb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 11:17:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/9/check/gate-nova-pep8/c35ff15/ : SUCCESS in 5m 44s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/9/check/gate-nova-docs/3f9060e//doc/build/html/ : SUCCESS in 4m 20s\n- gate-nova-python27 http://logs.openstack.org/18/274318/9/check/gate-nova-python27/9d2d421/ : SUCCESS in 10m 32s\n- gate-nova-python34 http://logs.openstack.org/18/274318/9/check/gate-nova-python34/adcf030/ : SUCCESS in 26m 21s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-full/55276b3/ : SUCCESS in 38m 47s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-postgres-full/ae66b96/ : SUCCESS in 43m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-neutron-full/abed3ac/ : SUCCESS in 56m 33s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/9/check/gate-grenade-dsvm/d375970/ : SUCCESS in 27m 52s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/9/check/gate-nova-tox-functional/42a6a10/ : SUCCESS in 10m 35s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/9/check/gate-grenade-dsvm-multinode/0426a19/ : SUCCESS in 45m 52s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/7d77922/ : SUCCESS in 32m 58s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-cells/27396a7/ : SUCCESS in 28m 12s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-full-ceph/1efa2ff/ : SUCCESS in 46m 08s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/23b01f2/ : SUCCESS in 49m 20s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/9/check/gate-tempest-dsvm-multinode-full/fd4d95f/ : SUCCESS in 1h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bf03dffe5d7f2fb70416f6715556fd185459a6ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-07 11:18:08.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/9/check/check-ibm-tempest-dsvm-full/669c842/ : SUCCESS in 1h 07m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"856d363fdb0b5ea9a293cf251d4591ff3be46f1f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-07 11:29:41.000000000","message":"Patch Set 9: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/9/8447/index.html : SUCCESS in 01 h 16 m 31 s","accounts_in_message":[],"_revision_number":9},{"id":"e4cc82a9a9cde5936c2b0dedb381ef2589eb1c78","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 11:34:16.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/9/ : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":9},{"id":"3e84a8b0761bb52dd863ea1aacb62ffaa0272960","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-07 11:38:10.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/078ec3c8072c424ea78ff381bea9833f/index.html : SUCCESS in 12m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1a8a851daef2401192815647f090fb18/index.html : SUCCESS in 17m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f7f64add72ea472bbca07ec0fec601b6/index.html : SUCCESS in 15m 31s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/9/check/real-db-upgrade_nova_percona_user_001:th-percona/27612edeb3804d0bb8a69338b6eaecb2/index.html : SUCCESS in 17m 56s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a8ceefcff35949dba171e35284f2e5eb/index.html : SUCCESS in 19m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/9/check/real-db-upgrade_nova_percona_user_002:th-percona/f2fef7e9fb214ab7a955188e7251d7e2/index.html : SUCCESS in 16m 39s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"2f5eaada1178451cf2db78602876cf4ed2ee7302","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-07 11:49:27.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/274318/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"39e73ef4b232d0cfadbf54404564410f0aea7556","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-07 13:25:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/9/check/dsvm-tempest-xen/04ce97c : SUCCESS in 1h 28m 46s","accounts_in_message":[],"_revision_number":9},{"id":"6de8fc14cb71e513e037672293cc301453cc6aa2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 17:43:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/9/check/check-dsvm-tempest-vz7-exe-minimal/cff676d : SUCCESS in 1h 04m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"361b0d8e86d4f01c0907f2b0a7306f9451ed786d","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-03-07 18:43:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"876b4a0974f9f3ee789d731cca3e92213f43afe1","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-03-07 18:49:04.000000000","message":"Patch Set 9:\n\n(2 comments)\n\nDone -- Added a little more info to the logging per your review.","accounts_in_message":[],"_revision_number":9},{"id":"4bf2ae438f3ef7b7baa909c68dcee75ae02e9bb6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 19:47:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/18/274318/10/check/gate-nova-pep8/1d238f1/ : SUCCESS in 2m 34s\n- gate-nova-docs http://docs-draft.openstack.org/18/274318/10/check/gate-nova-docs/31601e4//doc/build/html/ : SUCCESS in 9m 23s\n- gate-nova-python27 http://logs.openstack.org/18/274318/10/check/gate-nova-python27/0b18ea7/ : SUCCESS in 10m 39s\n- gate-nova-python34 http://logs.openstack.org/18/274318/10/check/gate-nova-python34/09abbc1/ : SUCCESS in 26m 34s\n- gate-tempest-dsvm-full http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-full/1a4573c/ : SUCCESS in 37m 48s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-postgres-full/8321a6e/ : SUCCESS in 38m 19s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-neutron-full/6aaefe4/ : SUCCESS in 1h 02m 29s\n- gate-grenade-dsvm http://logs.openstack.org/18/274318/10/check/gate-grenade-dsvm/ff461bd/ : SUCCESS in 48m 19s\n- gate-nova-tox-functional http://logs.openstack.org/18/274318/10/check/gate-nova-tox-functional/b626f2e/ : SUCCESS in 10m 41s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/274318/10/check/gate-grenade-dsvm-multinode/4a5bfaf/ : SUCCESS in 42m 44s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/52c4fd2/ : SUCCESS in 33m 39s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-cells/260e802/ : SUCCESS in 34m 07s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-full-ceph/4d98fed/ : SUCCESS in 38m 53s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/2e8cbbf/ : SUCCESS in 43m 02s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/18/274318/10/check/gate-tempest-dsvm-multinode-full/92f96ab/ : FAILURE in 46m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"108a612fdfe8b585e73542a3427e3d8906968e92","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-07 19:59:18.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/18/274318/10/check/check-ibm-tempest-dsvm-full/697d588/ : SUCCESS in 1h 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9f588ff32a0fee7634d5048de07aa94f1cc4b3ae","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-07 20:01:42.000000000","message":"Patch Set 10: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/18/274318/10/8547/index.html : SUCCESS in 01 h 12 m 24 s","accounts_in_message":[],"_revision_number":10},{"id":"de321a9bc6142fb303889c503c9db2b088ad34b5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-07 20:38:58.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/a8a4014a6d914235bd281e6394bc4923/index.html : SUCCESS in 15m 36s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8a4a2b90bf924ecdaba7f12317d6cb26/index.html : SUCCESS in 17m 11s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b5baac1d96cf467baed1593e02587776/index.html : SUCCESS in 13m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/10/check/real-db-upgrade_nova_percona_user_001:th-percona/bc57991952ee4da78f1436ddf3931e90/index.html : SUCCESS in 17m 48s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/18/274318/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/db9c35a140b74a8692255ed75982afbe/index.html : SUCCESS in 12m 55s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/18/274318/10/check/real-db-upgrade_nova_percona_user_002:th-percona/9e81f91833f74a19bfd4ee95ce4ddef0/index.html : SUCCESS in 16m 01s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"10aafde24bf8b811be0daf90e18da7eb0ce02fe6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-07 20:59:47.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/274318/10 : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"75154b4e9b91fbd5dca57019ef1e78a1c79dfccf","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-07 22:19:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/18/274318/10/check/dsvm-tempest-xen/d7da6c1 : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":10},{"id":"f5489d8a75454f49c0567dca388f5dd75911a3d3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 22:52:14.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274318/10/ : FAILURE in 3h 07m 39s","accounts_in_message":[],"_revision_number":10},{"id":"b4daeb589a80e437dd88a508fca392a283b93242","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-08 06:13:44.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/274318/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"67a58f303cc8cafb9d1664e9889e72d6445cb739","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-03-08 09:36:22.000000000","message":"Patch Set 10:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/18/274318/10 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":10},{"id":"32dc14d6466382372177c58a02e7bc4c56e180b9","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-08 12:46:29.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/18/274318/10/check/check-dsvm-tempest-vz7-exe-minimal/18bda7d : SUCCESS in 2h 20m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"738386144d2c362da89919dafd4367c7daead04f","author":{"_account_id":1865,"name":"Andrea Rosa","email":"andrea.rosa@gmail.com","username":"andrea-rosa-m"},"date":"2016-03-08 16:22:33.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"802a8aabdc78120383ceb39bdeebf3c32e7120aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-13 18:41:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/274318/10/2870 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"96e7714fa742650e1d79363f3d0ebf226e146777","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-05-10 21:08:22.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nFor the record, I\u0027m not crazy about nova stuffing things into the connector that gets passed back to cinder and then some, but not all, volume drivers can possibly use that to workaround the bug. It\u0027s technically an API change, although since it\u0027s additive it\u0027s not the worth thing ever - older cinder code won\u0027t know to check for it, and newer cinder code would have to be written under the assumption it might not be there (if talking to older nova before this change).\n\nscottda and hemna are looking at issues like this between nova and cinder and thinking about new APIs in cinder that nova can use for stuff like this, i.e. storing the connector in the cinder db with the volume attachment. But that would be an API change which we wouldn\u0027t backport.","accounts_in_message":[],"_revision_number":10},{"id":"2aed31b72351fa372a5d4544ff5f2cf99ecfeba8","author":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"date":"2016-05-10 22:45:03.000000000","message":"Abandoned\n\nCinder folks don\u0027t like this workaround either. No point polluting nova if there isn\u0027t significant cinder driver support.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e1d9c8200bcdb51fcc7970495c5b2a8475f2053c","revisions":{"d142f8269c3c07ba02b303f4265212370ba7df3d":{"kind":"REWORK","_number":1,"created":"2016-01-30 05:15:24.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/1"}}},"commit":{"parents":[{"commit":"26842bc9e70b600ebe4222c20049940eab5c8d1c","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26842bc9e70b600ebe4222c20049940eab5c8d1c"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to evacuate terminate connector.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d142f8269c3c07ba02b303f4265212370ba7df3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d142f8269c3c07ba02b303f4265212370ba7df3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"26842bc9e70b600ebe4222c20049940eab5c8d1c","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d1b564f3b91c7d6c9aa1dd6f13057b4fa5828bc8":{"kind":"REWORK","_number":2,"created":"2016-01-30 07:41:34.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/2"}}},"commit":{"parents":[{"commit":"2094bc3cd3fca8df200a97c22a82d08f1963d030","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2094bc3cd3fca8df200a97c22a82d08f1963d030"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 07:31:58.000000000","tz":-480},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to evacuate terminate connector.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1b564f3b91c7d6c9aa1dd6f13057b4fa5828bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1b564f3b91c7d6c9aa1dd6f13057b4fa5828bc8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2094bc3cd3fca8df200a97c22a82d08f1963d030","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7b177ccbb5c572e9b709089a6631ee1df0175e48":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-02-03 21:29:36.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/3"}}},"commit":{"parents":[{"commit":"434792a27b93f3b8fc79703a5cb9ada2de9174b0","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/434792a27b93f3b8fc79703a5cb9ada2de9174b0"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"Mark Sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-02-03 21:29:34.000000000","tz":0},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to evacuate terminate connector.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b177ccbb5c572e9b709089a6631ee1df0175e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b177ccbb5c572e9b709089a6631ee1df0175e48"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"434792a27b93f3b8fc79703a5cb9ada2de9174b0","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a46ef601e46c4947dfa730ba307cf03bd9713639":{"kind":"REWORK","_number":4,"created":"2016-02-06 05:54:36.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/4"}}},"commit":{"parents":[{"commit":"08ff7b101494e311f88fffd7aeea26dbf6f8568c","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08ff7b101494e311f88fffd7aeea26dbf6f8568c"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-02-06 05:49:34.000000000","tz":-480},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to evacuate terminate connector.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a46ef601e46c4947dfa730ba307cf03bd9713639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a46ef601e46c4947dfa730ba307cf03bd9713639"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"08ff7b101494e311f88fffd7aeea26dbf6f8568c","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d7a7c2b27df8384ff7f30531ed376832fb094225":{"kind":"REWORK","_number":5,"created":"2016-02-08 17:47:36.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/5"}}},"commit":{"parents":[{"commit":"08ff7b101494e311f88fffd7aeea26dbf6f8568c","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08ff7b101494e311f88fffd7aeea26dbf6f8568c"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-02-08 17:44:46.000000000","tz":-480},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to evacuate terminate connector.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d7a7c2b27df8384ff7f30531ed376832fb094225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d7a7c2b27df8384ff7f30531ed376832fb094225"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"08ff7b101494e311f88fffd7aeea26dbf6f8568c","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"757107525290afe902a2101fc0605c746b817c3d":{"kind":"NO_CHANGE","_number":6,"created":"2016-02-12 03:25:41.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/6"}}},"commit":{"parents":[{"commit":"9a248c954b23ecbe1853e3f291a87630cb42040c","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9a248c954b23ecbe1853e3f291a87630cb42040c"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"Mark Sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-02-12 03:25:40.000000000","tz":0},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to evacuate terminate connector.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/757107525290afe902a2101fc0605c746b817c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/757107525290afe902a2101fc0605c746b817c3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9a248c954b23ecbe1853e3f291a87630cb42040c","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6cf62881fb6706ec57ac4844b970bf82cc68ed0a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2016-02-12 04:47:41.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/7"}}},"commit":{"parents":[{"commit":"9a248c954b23ecbe1853e3f291a87630cb42040c","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9a248c954b23ecbe1853e3f291a87630cb42040c"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"Mark Sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-02-12 04:47:12.000000000","tz":0},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to the connector for evacuation.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6cf62881fb6706ec57ac4844b970bf82cc68ed0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6cf62881fb6706ec57ac4844b970bf82cc68ed0a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9a248c954b23ecbe1853e3f291a87630cb42040c","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4139923b77b741b32c8cd23881d78f2ec0e74f48":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2016-03-01 22:00:51.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/8"}}},"commit":{"parents":[{"commit":"1f2d2d46d576d12563512cd964b4b332b3a6706f","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f2d2d46d576d12563512cd964b4b332b3a6706f"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"Mark Sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-03-01 22:00:50.000000000","tz":0},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to the connector for evacuation.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4139923b77b741b32c8cd23881d78f2ec0e74f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4139923b77b741b32c8cd23881d78f2ec0e74f48"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1f2d2d46d576d12563512cd964b4b332b3a6706f","is_merged_in_target_branch":false,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"255d7db2b4cc78c9af2c761f97f52d00af2fbec3":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2016-03-07 10:10:02.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/18/274318/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/9"}}},"commit":{"parents":[{"commit":"a686185fc02ec421fd27270a343c19f668b95da6","subject":"Provide correct connector for evacuate terminate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a686185fc02ec421fd27270a343c19f668b95da6"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2016-03-07 10:10:00.000000000","tz":0},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to the connector for evacuation.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/255d7db2b4cc78c9af2c761f97f52d00af2fbec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/255d7db2b4cc78c9af2c761f97f52d00af2fbec3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a686185fc02ec421fd27270a343c19f668b95da6","is_merged_in_target_branch":true,"change_id":"I793f2996fc0af1c321a240ad9348dc9bce816030","change_number":266095,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e1d9c8200bcdb51fcc7970495c5b2a8475f2053c":{"kind":"REWORK","_number":10,"created":"2016-03-07 18:43:16.000000000","uploader":{"_account_id":11047,"name":"Mark Sturdevant","email":"mark.sturdevant@ibm.com","username":"markstur"},"ref":"refs/changes/18/274318/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/274318/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/274318/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/274318/10"}}},"commit":{"parents":[{"commit":"9517315ef20e30efa28348c83e5fa2e146bdbb82","subject":"Merge \"Remove unused methods in nova/utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9517315ef20e30efa28348c83e5fa2e146bdbb82"}]}],"author":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-01-30 04:29:46.000000000","tz":-480},"committer":{"name":"mark.sturdevant","email":"mark.sturdevant@hpe.com","date":"2016-03-07 18:42:50.000000000","tz":-480},"subject":"Add instance host to evacuate terminate connector","message":"Add instance host to evacuate terminate connector\n\nWhen we only have the connector for the wrong host,\nadd instance host to the connector for evacuation.\n\nWhen calling terminate_connection for evacuate, the\nconnector should be for the original host and not\nthe new host. Patch [1] provides a fix by using\nthe connector from init_connection which is stashed\nin bdm.connection_info, but that approach cannot fix\nthis bug for volumes that were attached before that\npatch.\n\nTo allow a fix for existing attachments, add the\ninstance host in connector[\u0027instance_host\u0027]. This\nwill allow cinder drivers to be patched to check\nfor this defect and use the instance host and volume\nto provide a workaround (the rest of the bogus\ncurrent host connector is left as-is to avoid creating\nnew defects such as validation or type errors).\n\nFor example:\n\n* Using host and volume a cinder driver may be able\nto discover the initiator or wwpns that are needed, or\n\n* Using the instance host instead of the connector host,\nsome drivers may not need the other parts of the connnector, or\n\n* Detecting the mismatch would allow some drivers to\nraise an exception instead of just trying to disconnect\nthe wrong host connection which might succeed or fail or\nleave dangling connections.\n\n[1] I793f2996fc0af1c321a240ad9348dc9bce816030\n\nChange-Id: I822e31bf8a739588991fcca34976d55b71b8a227\nPartial-Bug: #1522496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1d9c8200bcdb51fcc7970495c5b2a8475f2053c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1d9c8200bcdb51fcc7970495c5b2a8475f2053c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9517315ef20e30efa28348c83e5fa2e146bdbb82","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
