)]}'
{"id":"openstack%2Fnova~274363","triplet_id":"openstack%2Fnova~master~Ia4b591da8b8fbec64a261600c5f698d72d1f739a","project":"openstack/nova","branch":"master","hashtags":[],"change_id":"Ia4b591da8b8fbec64a261600c5f698d72d1f739a","subject":"[WIP] Switching on oslo.cache configuration options","status":"ABANDONED","created":"2016-01-30 17:47:35.000000000","updated":"2016-02-01 00:51:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c044af24fc5b6c4fcbe37cc84a154632f10b547a","_number":274363,"virtual_id_number":274363,"owner":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-01-31 23:07:55.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"date":"2016-01-31 22:30:48.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-01-31 22:44:00.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-01-31 22:02:54.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2016-01-31 22:41:19.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2016-01-31 22:29:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"date":"2016-01-31 21:44:41.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"date":"2016-01-31 22:16:29.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-01-31 21:43:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"all":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-01-31 21:41:04.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-01-31 04:07:00.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 04:21:39.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 21:43:21.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 21:44:41.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 22:02:54.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 22:16:29.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 22:29:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 22:30:48.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 22:41:19.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-01-31 22:44:00.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-01-31 23:07:55.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2622b2e87b7552dea4ac43903ac36952bcb71aaf","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-30 17:47:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0de3c0744174079e4e8dd85c35f7a21fa496560","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 18:04:39.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/274363/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6b7f17bea024515228e45824cfb735ccc012b225","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-30 18:21:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274363/1/Nova-ML2-Sriov : SUCCESS in 32m 51s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"203301c6126b56f842a12241262eec0f1497183c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 18:38:20.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"37489ebe895a1b7028af565354635e9d69b3b970","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-30 18:49:55.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/63/274363/1/check/check-ibm-tempest-dsvm-full/782ddec/ : SUCCESS in 1h 01m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e68c4de29fad29594fb4c74e863aef98756dd79c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 18:51:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/63/274363/1/check/check-dsvm-tempest-vz7-exe-minimal/2555f51 : SUCCESS in 1h 02m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"107b5a4d67f4ba1cc003d5ee7f700ce2ef7e91fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-30 18:55:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/63/274363/1/check/gate-nova-pep8/ed7a29d/ : SUCCESS in 8m 01s\n- gate-nova-docs http://docs-draft.openstack.org/63/274363/1/check/gate-nova-docs/ba3171b//doc/build/html/ : SUCCESS in 12m 34s\n- gate-nova-python27 http://logs.openstack.org/63/274363/1/check/gate-nova-python27/762bcaa/ : FAILURE in 15m 05s\n- gate-nova-python34 http://logs.openstack.org/63/274363/1/check/gate-nova-python34/3d4332a/ : FAILURE in 26m 45s\n- gate-nova-requirements http://logs.openstack.org/63/274363/1/check/gate-nova-requirements/1cebc62/ : SUCCESS in 1m 11s\n- gate-tempest-dsvm-full http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-full/c8e0f69/ : SUCCESS in 44m 41s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-postgres-full/c7cf0d7/ : SUCCESS in 44m 05s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-neutron-full/cbdea92/ : SUCCESS in 56m 29s\n- gate-grenade-dsvm http://logs.openstack.org/63/274363/1/check/gate-grenade-dsvm/3abd87a/ : SUCCESS in 48m 15s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-large-ops/69ea92b/ : SUCCESS in 31m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-neutron-large-ops/7756175/ : SUCCESS in 30m 46s\n- gate-nova-releasenotes http://docs-draft.openstack.org/63/274363/1/check/gate-nova-releasenotes/68d5adc//releasenotes/build/html/ : SUCCESS in 7m 27s\n- gate-nova-tox-functional http://logs.openstack.org/63/274363/1/check/gate-nova-tox-functional/6e8ec54/ : SUCCESS in 15m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/63/274363/1/check/gate-grenade-dsvm-multinode/7a19103/ : FAILURE in 47m 29s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/4576d4b/ : FAILURE in 31m 57s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-nova-v20-api/5226cbd/ : SUCCESS in 38m 34s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-nova-v20-api-legacy/15d2bf6/ : SUCCESS in 35m 59s\n- gate-tempest-dsvm-cells http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-cells/ae076da/ : SUCCESS in 39m 16s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-full-ceph/1574623/ : SUCCESS in 45m 41s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/e7601b8/ : SUCCESS in 48m 12s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/63/274363/1/check/gate-tempest-dsvm-multinode-full/ae766bf/ : SUCCESS in 1h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a979c60db3691cdfb6e81b07248358f00832a819","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-30 19:07:08.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/63/274363/1/4742/index.html : SUCCESS in 01 h 15 m 46 s","accounts_in_message":[],"_revision_number":1},{"id":"dbd070c0e3cd8332980cc376b839559ec4b7ec6a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-30 19:10:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/63/274363/1/check/dsvm-tempest-xen/6b82749 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0be2b33c70282d3ed04fe97218a7e6ab5a1ebca7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 19:31:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274363/1/ : SUCCESS in 1h 42m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e917011e7c5af6aab401f2d8187cd4bd5a64635c","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-30 20:23:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80e9dc085529f50ec86f5ba0c13241af757aa08a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 20:40:42.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/274363/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"df52f5c5e79974b0dc1286a63a0b563665d06a83","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 21:13:47.000000000","message":"Patch Set 2:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/2 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"e2ac555876b6e54cf069fa3cac72819e3327ae75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-30 21:28:10.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/63/274363/2/check/check-ibm-tempest-dsvm-full/691c7c6/ : SUCCESS in 1h 03m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24aceaa25bf2dea8e535443fb2a0e4890e4efe00","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 21:32:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/63/274363/2/check/check-dsvm-tempest-vz7-exe-minimal/769a14c : SUCCESS in 1h 08m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"8efdc2500834e42a8036e0968bba9206d7f7d468","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-30 21:39:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/63/274363/2/check/gate-nova-pep8/6db5b38/ : SUCCESS in 8m 35s\n- gate-nova-docs http://docs-draft.openstack.org/63/274363/2/check/gate-nova-docs/a1987cb//doc/build/html/ : SUCCESS in 13m 50s\n- gate-nova-python27 http://logs.openstack.org/63/274363/2/check/gate-nova-python27/904f0c7/ : SUCCESS in 15m 16s\n- gate-nova-python34 http://logs.openstack.org/63/274363/2/check/gate-nova-python34/a7eef4a/ : SUCCESS in 27m 09s\n- gate-nova-requirements http://logs.openstack.org/63/274363/2/check/gate-nova-requirements/29402d2/ : SUCCESS in 1m 10s\n- gate-tempest-dsvm-full http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-full/d39105c/ : SUCCESS in 45m 09s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-postgres-full/05267e5/ : SUCCESS in 42m 49s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-neutron-full/8d496ea/ : SUCCESS in 59m 21s\n- gate-grenade-dsvm http://logs.openstack.org/63/274363/2/check/gate-grenade-dsvm/cf18b66/ : SUCCESS in 44m 11s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-large-ops/27f69f4/ : SUCCESS in 35m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-neutron-large-ops/4ffb8ff/ : SUCCESS in 30m 33s\n- gate-nova-releasenotes http://docs-draft.openstack.org/63/274363/2/check/gate-nova-releasenotes/59db262//releasenotes/build/html/ : SUCCESS in 7m 47s\n- gate-nova-tox-functional http://logs.openstack.org/63/274363/2/check/gate-nova-tox-functional/d1a5d74/ : SUCCESS in 15m 55s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/63/274363/2/check/gate-grenade-dsvm-multinode/f852c26/ : FAILURE in 45m 51s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3cfdbcf/ : FAILURE in 29m 57s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-nova-v20-api/894c3f1/ : SUCCESS in 40m 47s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-nova-v20-api-legacy/53af608/ : SUCCESS in 36m 59s\n- gate-tempest-dsvm-cells http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-cells/4bd84c3/ : SUCCESS in 40m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-full-ceph/1b4e77d/ : SUCCESS in 47m 45s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/a0a4b79/ : SUCCESS in 50m 47s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/63/274363/2/check/gate-tempest-dsvm-multinode-full/4bb71d6/ : FAILURE in 1h 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8e7ba2982dcac7f4bf38083210beb3ea7498d9b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-30 21:39:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/63/274363/2/check/dsvm-tempest-xen/f987185 : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b4369b61c62868c1af8ec01084c508b4e0e096c2","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-30 21:43:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"03a6ebd70208c04edd9cf771448f09c75c49745c","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-30 21:47:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d9596d29016b6b247306cf8403ac301b544b5f04","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-01-30 21:47:36.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-274363-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"f85594d3ff83895e565e5bdbba7e3890a0d5c72f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 22:00:43.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/274363/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"dc12f618add9d8688b647d82e3b2bed91c8f034d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 22:04:22.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/274363/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6c19eaa371c99fff28d7b6e131b83e94f1a65d0f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 22:33:18.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"1702130c2ac663d58ff552a8ea5611ae706f9386","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 22:39:16.000000000","message":"Patch Set 4:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/4 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":4},{"id":"7788b2ac6670ca208ba076345b53883e097ae88d","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-30 22:42:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d58fdfddc12076ddd076c9e36849d65244814d12","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-30 23:00:10.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/274363/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e926a998836ec17183453a526de2e79963cf7d09","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 23:32:35.000000000","message":"Patch Set 5:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/5 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"5ad47f1bcd98c7eee0bb73e208b2b3592967f602","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-30 23:46:30.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/63/274363/5/check/check-ibm-tempest-dsvm-full/a0f949d/ : SUCCESS in 1h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"22b3d354277efcf4b1856030d29c02776e428e2e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-30 23:56:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/63/274363/5/check/check-dsvm-tempest-vz7-exe-minimal/10a2817 : SUCCESS in 1h 11m 41s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"4b0afb725e74e871e97957ef5f7e75e845a4f2df","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-30 23:58:52.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/63/274363/5/4746/index.html : SUCCESS in 01 h 12 m 47 s","accounts_in_message":[],"_revision_number":5},{"id":"579b6e831cce90d0368a62d348218a11a03b1d72","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-30 23:59:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/63/274363/5/check/dsvm-tempest-xen/86e9096 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":5},{"id":"540a312cc0f62f50828bde873b24f3ae28f9fd89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-31 00:00:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/63/274363/5/check/gate-nova-pep8/f4c6b4f/ : SUCCESS in 9m 08s\n- gate-nova-docs http://docs-draft.openstack.org/63/274363/5/check/gate-nova-docs/808a9e0//doc/build/html/ : SUCCESS in 13m 05s\n- gate-nova-python27 http://logs.openstack.org/63/274363/5/check/gate-nova-python27/cb40e3f/ : SUCCESS in 19m 00s\n- gate-nova-python34 http://logs.openstack.org/63/274363/5/check/gate-nova-python34/3d10141/ : SUCCESS in 27m 43s\n- gate-nova-requirements http://logs.openstack.org/63/274363/5/check/gate-nova-requirements/29a3493/ : SUCCESS in 1m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-full/8939a19/ : SUCCESS in 53m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-postgres-full/647ed58/ : SUCCESS in 52m 55s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-neutron-full/a5c4697/ : SUCCESS in 57m 30s\n- gate-grenade-dsvm http://logs.openstack.org/63/274363/5/check/gate-grenade-dsvm/f07a080/ : SUCCESS in 44m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-large-ops/2a4990a/ : SUCCESS in 31m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-neutron-large-ops/e6aba15/ : SUCCESS in 31m 46s\n- gate-nova-releasenotes http://docs-draft.openstack.org/63/274363/5/check/gate-nova-releasenotes/8fb40e2//releasenotes/build/html/ : SUCCESS in 7m 54s\n- gate-nova-tox-functional http://logs.openstack.org/63/274363/5/check/gate-nova-tox-functional/e5f05c3/ : SUCCESS in 17m 06s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/63/274363/5/check/gate-grenade-dsvm-multinode/7107519/ : SUCCESS in 49m 49s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/47a93d3/ : FAILURE in 32m 00s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-nova-v20-api/7d3d80a/ : SUCCESS in 39m 47s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-nova-v20-api-legacy/dddc0ea/ : SUCCESS in 39m 50s\n- gate-tempest-dsvm-cells http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-cells/42e846c/ : SUCCESS in 42m 45s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-full-ceph/26ab54f/ : SUCCESS in 48m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/64c99cb/ : SUCCESS in 38m 20s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/63/274363/5/check/gate-tempest-dsvm-multinode-full/179f785/ : FAILURE in 1h 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67618da5da24c1ac0ce96d2f0c595f1992830eed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-01-31 00:13:00.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-274363-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"d445e285698659a80150e0692970d066a8b1422c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-31 01:29:11.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/59001a8593484db3b89815121e911035/index.html : SUCCESS in 13m 51s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b39b3038c1e349c0a6ce2c3bdae90181/index.html : SUCCESS in 15m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0507d330813849b0b2adee0dcbd07c2a/index.html : SUCCESS in 14m 12s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/5/check/real-db-upgrade_nova_percona_user_001:th-percona/bb7c8885510a468082fca1ddb463bef7/index.html : SUCCESS in 12m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/45d43327b5824c11bbe25c23743d1ed4/index.html : SUCCESS in 14m 47s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/5/check/real-db-upgrade_nova_percona_user_002:th-percona/dba2505d66524ad0b2618708dbfc4f88/index.html : SUCCESS in 13m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"fd1af9ce50e8a1d2294aba66965dbc0ff8affcba","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 02:41:10.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274363/5/ : FAILURE in 3h 57m 58s","accounts_in_message":[],"_revision_number":5},{"id":"fcb5a14ae526500950f7c7239d534133be5e5065","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-31 02:52:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3ce4d10c42df5afd75da1288bff9798e0443f0f7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-31 03:09:52.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/274363/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"0c1de12927934e8ef2360ba61e18e03abcd41ddf","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-31 03:13:26.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d320b0decf6b4589a027f8c6d2fc2539/index.html : SUCCESS in 12m 45s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/21d6162d520144cbbb0c272115db8ef3/index.html : SUCCESS in 14m 24s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/4a8f43e17599418abe0823eb6a8ce83e/index.html : SUCCESS in 13m 51s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/6/check/real-db-upgrade_nova_percona_user_001:th-percona/14f8c1823c4e417f898a712f465d6fb7/index.html : SUCCESS in 15m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/250f02912e0b4e7fa3af072c5fc13b26/index.html : SUCCESS in 13m 59s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/6/check/real-db-upgrade_nova_percona_user_002:th-percona/ddceebb5bba3442ab670497f9abc8eab/index.html : SUCCESS in 14m 22s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"bdb150dcf6ef01968ab4bd3e35c5a9debafc8ee9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 03:45:32.000000000","message":"Patch Set 6:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/6 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":6},{"id":"a153c1fc8eab393fff14f76b3a0e6c346865085e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 03:55:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/63/274363/6/check/check-dsvm-tempest-vz7-exe-minimal/630f18b : SUCCESS in 1h 02m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"e68aa66f9c5bacb142ca8c9a6eb6172943001700","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-31 03:56:37.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/63/274363/6/check/check-ibm-tempest-dsvm-full/2a94315/ : SUCCESS in 1h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce39f50cc9d9e850e230ca6d224424faf0e393f3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-01-31 04:07:00.000000000","message":"Patch Set 6: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/63/274363/6/4749/index.html : SUCCESS in 01 h 11 m 42 s","accounts_in_message":[],"_revision_number":6},{"id":"385a718925fa907138655f3151ae90a8267a6347","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-31 04:11:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/63/274363/6/check/gate-nova-pep8/acb5b9c/ : SUCCESS in 7m 42s\n- gate-nova-docs http://docs-draft.openstack.org/63/274363/6/check/gate-nova-docs/1cd8519//doc/build/html/ : SUCCESS in 12m 24s\n- gate-nova-python27 http://logs.openstack.org/63/274363/6/check/gate-nova-python27/fb8cc9c/ : SUCCESS in 14m 19s\n- gate-nova-python34 http://logs.openstack.org/63/274363/6/check/gate-nova-python34/9dea68f/ : SUCCESS in 26m 31s\n- gate-nova-requirements http://logs.openstack.org/63/274363/6/check/gate-nova-requirements/8b25a5b/ : SUCCESS in 1m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-full/6b87a8d/ : SUCCESS in 45m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-postgres-full/3e000c3/ : SUCCESS in 47m 25s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-neutron-full/6134082/ : SUCCESS in 58m 51s\n- gate-grenade-dsvm http://logs.openstack.org/63/274363/6/check/gate-grenade-dsvm/53d7999/ : SUCCESS in 44m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-large-ops/47796a8/ : SUCCESS in 30m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-neutron-large-ops/772e9b7/ : SUCCESS in 32m 30s\n- gate-nova-releasenotes http://docs-draft.openstack.org/63/274363/6/check/gate-nova-releasenotes/717ed53//releasenotes/build/html/ : SUCCESS in 8m 16s\n- gate-nova-tox-functional http://logs.openstack.org/63/274363/6/check/gate-nova-tox-functional/c3a3381/ : SUCCESS in 15m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/63/274363/6/check/gate-grenade-dsvm-multinode/9029659/ : SUCCESS in 48m 04s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2f3dbf1/ : FAILURE in 28m 23s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-nova-v20-api/58536c5/ : SUCCESS in 37m 04s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-nova-v20-api-legacy/b0a0649/ : SUCCESS in 41m 57s\n- gate-tempest-dsvm-cells http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-cells/c694817/ : SUCCESS in 40m 10s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-full-ceph/10b438b/ : SUCCESS in 45m 03s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/f382d07/ : SUCCESS in 47m 34s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/63/274363/6/check/gate-tempest-dsvm-multinode-full/6ee346d/ : FAILURE in 1h 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"95c527c4f2e796e511bded17e2e9bb3b6af8f434","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-31 04:13:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/63/274363/6/check/dsvm-tempest-xen/095b4b0 : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":6},{"id":"3e0de8004ecf847cf22d120577b7c9799fb11a40","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-01-31 04:21:39.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-274363-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"94b658dca95fecfb918d098756fe4fddb5c2a97e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 06:51:52.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274363/6/ : FAILURE in 3h 58m 22s","accounts_in_message":[],"_revision_number":6},{"id":"7a0ec391b23f4fcb641348c451f142f3c4511fc3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 07:18:12.000000000","message":"Patch Set 6:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":6},{"id":"080be7f1fad1f2a8befe361080c285f096208673","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 11:17:14.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274363/6/ : FAILURE in 3h 58m 15s","accounts_in_message":[],"_revision_number":6},{"id":"3a8582b94b4d04b6de4da4a0c477186982cb2958","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-31 21:26:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1abdf3ba3af274ee7376f653e77a012e6016d42c","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-01-31 21:41:04.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"de6eb90f90eb7786d666c929efdcfd1b91148849","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-01-31 21:43:21.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/fd0c631d2a0e4528a01d9fee59333dbe/index.html : SUCCESS in 13m 51s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8823ce20f919427c86d65506dfbbc327/index.html : SUCCESS in 15m 26s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/bdfe826228dc423b8d14b4fe36f86127/index.html : SUCCESS in 12m 19s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/7/check/real-db-upgrade_nova_percona_user_001:th-percona/56b2ad3f7a624e71b0d92ecbeb53bec6/index.html : SUCCESS in 15m 39s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/63/274363/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a66aba8c0c5f42ddaf99e6805e443b6a/index.html : SUCCESS in 11m 56s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/63/274363/7/check/real-db-upgrade_nova_percona_user_002:th-percona/12df2e34426e433aa4ed3698e80e757f/index.html : SUCCESS in 12m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"b24d795780cae37cafa6ff78909f65e38b8eec50","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-01-31 21:44:41.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/274363/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"5e1e3a4880d77595a78072915cf6c91b2d96edf9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-01-31 22:02:54.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/274363/7/Nova-ML2-Sriov : FAILURE in 35m 09s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"aaa123e83748d40590b7bd9800e0e000a477579f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 22:16:29.000000000","message":"Patch Set 7:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/63/274363/7 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":7},{"id":"ceb80e7f1ac4a412d813474107918ada7b99645b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-01-31 22:29:55.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/63/274363/7/check/check-ibm-tempest-dsvm-full/b01d039/ : SUCCESS in 1h 02m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2499cc0a977c799b801f64c8c5c0bbf1d6dbe145","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 22:30:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/63/274363/7/check/check-dsvm-tempest-vz7-exe-minimal/166e702 : SUCCESS in 1h 03m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"0b927d2a3fdeab60104d93f8a43b7141d882041c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-01-31 22:41:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/63/274363/7/check/dsvm-tempest-xen/ae8ee8f : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":7},{"id":"ed17d5f9990a3d9b46fc4e63b639bc76e58ae14c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-31 22:44:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/63/274363/7/check/gate-nova-pep8/c09402a/ : SUCCESS in 7m 29s\n- gate-nova-docs http://logs.openstack.org/63/274363/7/check/gate-nova-docs/c579934/ : FAILURE in 7m 35s\n- gate-nova-python27 http://logs.openstack.org/63/274363/7/check/gate-nova-python27/9f1b07f/ : SUCCESS in 14m 08s\n- gate-nova-python34 http://logs.openstack.org/63/274363/7/check/gate-nova-python34/c90362e/ : SUCCESS in 29m 57s\n- gate-nova-requirements http://logs.openstack.org/63/274363/7/check/gate-nova-requirements/b379357/ : SUCCESS in 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-full/c8f4522/ : SUCCESS in 45m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-postgres-full/048d979/ : SUCCESS in 39m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-neutron-full/8b65fb9/ : SUCCESS in 1h 03m 17s\n- gate-grenade-dsvm http://logs.openstack.org/63/274363/7/check/gate-grenade-dsvm/aaa0239/ : SUCCESS in 47m 55s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-large-ops/3a930ab/ : SUCCESS in 31m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-neutron-large-ops/87353a0/ : SUCCESS in 31m 32s\n- gate-nova-releasenotes http://docs-draft.openstack.org/63/274363/7/check/gate-nova-releasenotes/a81ba24//releasenotes/build/html/ : SUCCESS in 7m 22s\n- gate-nova-tox-functional http://logs.openstack.org/63/274363/7/check/gate-nova-tox-functional/c6cdc81/ : SUCCESS in 15m 43s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/63/274363/7/check/gate-grenade-dsvm-multinode/4dd0de5/ : FAILURE in 45m 21s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/711b148/ : FAILURE in 36m 36s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-nova-v20-api/ef561d5/ : SUCCESS in 35m 52s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-nova-v20-api-legacy/7d98d69/ : SUCCESS in 36m 43s\n- gate-tempest-dsvm-cells http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-cells/d92f23f/ : SUCCESS in 40m 30s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-full-ceph/7f63bef/ : SUCCESS in 55m 09s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/e7114d4/ : FAILURE in 44m 51s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/63/274363/7/check/gate-tempest-dsvm-multinode-full/15d1cde/ : FAILURE in 1h 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0687a8e57608421859a6e7731cf254f72df356a2","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-01-31 23:07:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/274363/7/ : SUCCESS in 1h 40m 43s","accounts_in_message":[],"_revision_number":7},{"id":"c044af24fc5b6c4fcbe37cc84a154632f10b547a","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-02-01 00:51:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1d328df745bc61a557634da01a39e7a73099d494","revisions":{"2807e1ec2e8455dbcfcca78a96a2d94fc7fc0863":{"kind":"REWORK","_number":1,"created":"2016-01-30 17:47:35.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/1"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 17:46:23.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2807e1ec2e8455dbcfcca78a96a2d94fc7fc0863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2807e1ec2e8455dbcfcca78a96a2d94fc7fc0863"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"90b2e56cb3ae5c9e8ef5c730db1180c0d22ad76e":{"kind":"REWORK","_number":2,"created":"2016-01-30 20:23:27.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/2"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 20:22:47.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90b2e56cb3ae5c9e8ef5c730db1180c0d22ad76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90b2e56cb3ae5c9e8ef5c730db1180c0d22ad76e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"2bea8abed4e68c9dd86f1abadf3b8d19861a8f11":{"kind":"REWORK","_number":3,"created":"2016-01-30 21:43:00.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/3"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 21:41:38.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bea8abed4e68c9dd86f1abadf3b8d19861a8f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bea8abed4e68c9dd86f1abadf3b8d19861a8f11"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"39ccad8b4b5e76c4509b0794599fd053d053cd5e":{"kind":"REWORK","_number":4,"created":"2016-01-30 21:47:29.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/4"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 21:45:55.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39ccad8b4b5e76c4509b0794599fd053d053cd5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39ccad8b4b5e76c4509b0794599fd053d053cd5e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"c03552498511cfca8be200381a1af1f3268df667":{"kind":"REWORK","_number":5,"created":"2016-01-30 22:42:15.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/5"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 22:39:48.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c03552498511cfca8be200381a1af1f3268df667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c03552498511cfca8be200381a1af1f3268df667"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"41694b5e152ef4143f1074563caec52bae615784":{"kind":"REWORK","_number":6,"created":"2016-01-31 02:52:42.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/6"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-31 02:51:23.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/41694b5e152ef4143f1074563caec52bae615784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/41694b5e152ef4143f1074563caec52bae615784"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"1d328df745bc61a557634da01a39e7a73099d494":{"kind":"REWORK","_number":7,"created":"2016-01-31 21:26:23.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/63/274363/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/274363/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/274363/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/274363/7"}}},"commit":{"parents":[{"commit":"5cc05063c783a86862583dc334cb58a7b03e33f6","subject":"[WIP] Trying alternative patch for oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cc05063c783a86862583dc334cb58a7b03e33f6"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-30 16:10:08.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2016-01-31 21:25:32.000000000","tz":-300},"subject":"[WIP] Switching on oslo.cache configuration options","message":"[WIP] Switching on oslo.cache configuration options\n\nChange-Id: Ia4b591da8b8fbec64a261600c5f698d72d1f739a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d328df745bc61a557634da01a39e7a73099d494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d328df745bc61a557634da01a39e7a73099d494"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cc05063c783a86862583dc334cb58a7b03e33f6","is_merged_in_target_branch":false,"change_id":"I9bebaded652e818a8dbae0f58870a541d406ea4f","change_number":274180,"patch_set_number":8,"change_status":"ABANDONED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
