)]}'
{"id":"openstack%2Fnova~275624","triplet_id":"openstack%2Fnova~master~I39790799e3d278c96e022b6ec04c310bab3b6895","project":"openstack/nova","branch":"master","topic":"ovs_offload","hashtags":[],"change_id":"I39790799e3d278c96e022b6ec04c310bab3b6895","subject":"[WIP] OVS offload POC","status":"ABANDONED","created":"2016-02-03 11:02:20.000000000","updated":"2016-11-16 06:20:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92165f2c222e9deea20e78c265a34004c6b61fd9","_number":275624,"virtual_id_number":275624,"owner":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"disliked":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"date":"2016-11-14 12:23:01.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"date":"2016-11-14 12:08:45.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-11-14 12:06:33.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-11-14 12:45:21.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-11-14 12:21:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2016-11-14 11:30:34.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},{"value":1,"date":"2016-11-14 19:26:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2016-11-14 13:25:45.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-11-14 13:49:40.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-11-14 12:59:00.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"date":"2016-11-14 13:15:15.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},{"value":1,"date":"2016-11-14 14:39:44.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-21 08:50:13.000000000","updated_by":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-31 06:10:31.000000000","updated_by":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"reviewer":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"state":"REVIEWER"},{"updated":"2016-03-31 11:03:26.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2016-04-04 23:53:14.000000000","updated_by":{"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},"reviewer":{"_account_id":12237,"name":"Qijing Li","email":"qijing_li@symantec.com","username":"stack"},"state":"REVIEWER"},{"updated":"2016-06-02 10:42:39.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 11:30:34.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 12:06:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-14 12:08:45.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 12:21:35.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 12:23:01.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 12:45:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 12:59:00.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 13:15:15.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 13:25:45.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 13:49:40.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 14:39:44.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 19:26:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e5fcf710135372abb323868dc613dcdd18630a6","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-03 11:02:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7480dde22f13a4161ebecb774da84ecf3e069cc","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-03 11:03:04.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"bf3c96e43aa361abc04f8f5bf643be2b64918234","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-03 11:25:30.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/275624/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a0922d1d9136d0ef9869ca276bd9507140e18436","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-03 11:46:07.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9dcf4cfd6a324073a00210374122b1ee/index.html : SUCCESS in 12m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/004ce09a142b4d13a8379a1700ef2f82/index.html : SUCCESS in 12m 56s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1aa4645c269d457e979981f9f38c4846/index.html : SUCCESS in 14m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/1/check/real-db-upgrade_nova_percona_user_001:th-percona/931fb91edd294ebbba2c7baca8b406f4/index.html : SUCCESS in 15m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d8b294850bbf4e1d91db8dc451a3bd4c/index.html : SUCCESS in 16m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/1/check/real-db-upgrade_nova_percona_user_002:th-percona/dc44adb2071b4b25a30b4c162a467475/index.html : SUCCESS in 14m 08s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"d63bffcc8b4e5109951c0db0ac887ce86c4f2404","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-03 12:14:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/24/275624/1/check/gate-nova-pep8/b4fddc6/ : FAILURE in 8m 16s\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/1/check/gate-nova-docs/d18fb2a//doc/build/html/ : SUCCESS in 12m 12s\n- gate-nova-python27 http://logs.openstack.org/24/275624/1/check/gate-nova-python27/450c98b/ : SUCCESS in 15m 12s\n- gate-nova-python34 http://logs.openstack.org/24/275624/1/check/gate-nova-python34/de46cb3/ : SUCCESS in 27m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-full/1d2d7ad/ : SUCCESS in 45m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-postgres-full/054692d/ : SUCCESS in 43m 07s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-neutron-full/5808db1/ : FAILURE in 13m 12s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/1/check/gate-grenade-dsvm/7083f72/ : FAILURE in 15m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-large-ops/9c512fe/ : FAILURE in 12m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-neutron-large-ops/846c964/ : SUCCESS in 30m 17s\n- gate-nova-tox-functional http://logs.openstack.org/24/275624/1/check/gate-nova-tox-functional/11d20d1/ : SUCCESS in 16m 09s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/1/check/gate-grenade-dsvm-multinode/a2a2b98/ : SUCCESS in 51m 26s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a2102af/ : SUCCESS in 40m 35s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-nova-v20-api/098bcc9/ : FAILURE in 12m 46s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-nova-v20-api-legacy/9f0e530/ : FAILURE in 12m 57s\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-cells/e423317/ : SUCCESS in 40m 02s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-full-ceph/b0e52c3/ : SUCCESS in 52m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/cc36a07/ : SUCCESS in 52m 29s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/1/check/gate-tempest-dsvm-multinode-full/060e537/ : SUCCESS in 1h 07m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc01add4b54f679245b233ec74f28979599e181e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-03 12:18:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/1/check/dsvm-tempest-xen/57e1f1d : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"abe4fb184368b875b3d709f52fa890be95910523","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-03 12:21:25.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/1/5078/index.html : SUCCESS in 01 h 16 m 01 s","accounts_in_message":[],"_revision_number":1},{"id":"265ddf2d5cc74ad320957c90601af9cbba3415fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-03 13:08:03.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/1/check/check-ibm-tempest-dsvm-full/16a3c81/ : FAILURE in 2h 05m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"455d605d15be3017f3d0d6e9d555c0d92102af23","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-03 16:25:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/1/check/check-dsvm-tempest-vz7-exe-minimal/cd27cf2 : SUCCESS in 1h 57m 54s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"76845016035531cea04ff870ea1579ee522a2b49","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-03 17:32:00.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/1/ : FAILURE in 3h 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"64aeb4fb1d16467e761f4c347cc9b4e8e4426a62","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-03 21:10:09.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"516dd7990ca6f7cb19c2a65e5c3892eb44908341","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-04 12:55:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6855fb0a341d77cc91c7a6fb6c73d51a45cc2f5","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-04 13:00:43.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d5ab869134c7c6bb72ade7de9a5847b0f63c7839","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-04 13:02:26.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/275624/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"98a61d4327bdd9ca989609c9b77d79af4c76e3be","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-04 13:14:29.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/2/check/check-ibm-tempest-dsvm-full/d4fb6da/ : FAILURE in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a98b8d9b3cecb50a5ce7142f8b36528b86538020","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-04 13:23:06.000000000","message":"Patch Set 2: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/2/5268/index.html : FAILURE in 25 m 58 s","accounts_in_message":[],"_revision_number":2},{"id":"7568f6e0bb706eae5035f1449be03d18c36ac42a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-04 13:34:26.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8a1b56d652f0451a8cf0a41187c79687/index.html : SUCCESS in 12m 07s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/5df2f4f6aaf246199ee5dbbaaa7686cb/index.html : SUCCESS in 12m 48s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5c1f6a2fba4e4556bba2bf16217aac36/index.html : SUCCESS in 12m 14s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/2/check/real-db-upgrade_nova_percona_user_001:th-percona/384e8e8f79c74e3cad11d10906d32f51/index.html : SUCCESS in 16m 15s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/31a2f50cf8e244998dd5a4b5d89e8539/index.html : SUCCESS in 12m 41s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/2/check/real-db-upgrade_nova_percona_user_002:th-percona/87b16d1a41024990baf2406342eade5c/index.html : SUCCESS in 14m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"c55b96d89bebd5e1a987814322be7922ba818848","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-04 14:15:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/24/275624/2/check/gate-nova-pep8/292d569/ : FAILURE in 7m 37s\n- gate-nova-docs http://logs.openstack.org/24/275624/2/check/gate-nova-docs/74af294/ : FAILURE in 7m 40s\n- gate-nova-python27 http://logs.openstack.org/24/275624/2/check/gate-nova-python27/fd9dbca/ : FAILURE in 7m 43s\n- gate-nova-python34 http://logs.openstack.org/24/275624/2/check/gate-nova-python34/6a5c568/ : FAILURE in 9m 12s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-full/7bdd510/ : FAILURE in 20m 46s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-postgres-full/791aa84/ : FAILURE in 20m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-neutron-full/da7e40f/ : FAILURE in 22m 04s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/2/check/gate-grenade-dsvm/cae98a5/ : FAILURE in 44m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-large-ops/6acd0ed/ : FAILURE in 19m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-neutron-large-ops/1a29605/ : FAILURE in 20m 57s\n- gate-nova-tox-functional http://logs.openstack.org/24/275624/2/check/gate-nova-tox-functional/f1d4534/ : FAILURE in 7m 16s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/2/check/gate-grenade-dsvm-multinode/6e37a2c/ : FAILURE in 59m 47s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0bbda6c/ : FAILURE in 25m 38s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-nova-v20-api/8046679/ : FAILURE in 22m 10s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-nova-v20-api-legacy/6806f26/ : FAILURE in 21m 29s\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-cells/8f017f4/ : FAILURE in 21m 07s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-full-ceph/22d239d/ : FAILURE in 23m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/505b1cd/ : FAILURE in 23m 35s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/2/check/gate-tempest-dsvm-multinode-full/862d250/ : FAILURE in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc35be6ac3312de38227f721d3f7fa02844f32e3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 14:50:16.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/2/ : FAILURE in 1h 54m 13s","accounts_in_message":[],"_revision_number":2},{"id":"28083cec72752af9d49851974d09b14faff8b72d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 14:59:58.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/2/check/check-dsvm-tempest-vz7-exe-minimal/0b518db : FAILURE in 26m 39s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"72e4853bb39292299711ae885f09fbc49c2c781b","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-04 15:22:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cab1044531c661415407d69bca4e2c89d79c5c5e","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-04 15:25:56.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"6888a8a826d06e821d7fb1911c6a2c066ec08dd6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-04 15:45:01.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/275624/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"44581eebcc3de5ebf8f08e81cb5432a14cf25e1b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-04 16:08:06.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0e1742d5cbf14ad39e636d61384a784b/index.html : SUCCESS in 15m 10s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/bad23afca1fb45a8b4d0c12b21aa4143/index.html : SUCCESS in 22m 46s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9414449d09954f2dbc1368206df4c474/index.html : SUCCESS in 14m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/3/check/real-db-upgrade_nova_percona_user_001:th-percona/c5e31ca32df3448b8393ee26bc6ef59c/index.html : SUCCESS in 17m 34s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3dcf965bb95f496694d8b4615e22a69a/index.html : SUCCESS in 14m 37s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/3/check/real-db-upgrade_nova_percona_user_002:th-percona/832f617311224ea2bc40c5dda94635d2/index.html : SUCCESS in 15m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"c10484f556d9c90cd689b444f5ac5365b3dc681c","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-04 16:17:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee0abc41beb6db34dddb1467c2d5bad51c050f42","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-02-04 16:17:41.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"bc1916d7e44b74e4f09dc0aa6ca7dab1e48fb0ef","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-04 16:42:14.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/275624/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ba1de8e0e4f8722b74abaf6d8c6f3f565bdeac57","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-04 17:19:41.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/4/check/check-ibm-tempest-dsvm-full/674617e/ : SUCCESS in 1h 01m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c230082f01dd98f70b6281a93d1eca253086c7ef","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-04 17:45:38.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/4/5297/index.html : SUCCESS in 01 h 19 m 06 s","accounts_in_message":[],"_revision_number":4},{"id":"73fc8ebc5f2568b8dab30193e80f14f98170f853","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-04 17:53:00.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f080231d38c94d8991f8f31d8a7e4f6d/index.html : SUCCESS in 14m 07s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/830cd35cf6634b0cad09c92ae972ba66/index.html : SUCCESS in 1h 26m 51s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/59fa857c2d5e4cd883947b69eee2f0e5/index.html : SUCCESS in 15m 19s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/4/check/real-db-upgrade_nova_percona_user_001:th-percona/94ca5279de914c39b20cabd7310ae193/index.html : SUCCESS in 15m 03s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/b56ee0da26de492eb01fda71b31cb724/index.html : SUCCESS in 12m 39s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/4/check/real-db-upgrade_nova_percona_user_002:th-percona/1667e54106be45d4b2fef9bf8fab5ade/index.html : SUCCESS in 14m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"6ecba0ab98f1b143dd0c85077bac53b1e48cb703","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-04 19:44:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/24/275624/4/check/gate-nova-pep8/3e988bd/ : SUCCESS in 9m 38s\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/4/check/gate-nova-docs/2604af8//doc/build/html/ : SUCCESS in 14m 33s\n- gate-nova-python27 http://logs.openstack.org/24/275624/4/check/gate-nova-python27/b7039fd/ : SUCCESS in 14m 51s\n- gate-nova-python34 http://logs.openstack.org/24/275624/4/check/gate-nova-python34/1bd5cc8/ : SUCCESS in 29m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-full/8ebeb69/ : SUCCESS in 40m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-postgres-full/a7a3dc1/ : SUCCESS in 47m 03s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-neutron-full/f9dfe33/ : SUCCESS in 1h 00m 52s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/4/check/gate-grenade-dsvm/30a7de6/ : SUCCESS in 45m 13s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-large-ops/834bfe7/ : SUCCESS in 30m 11s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-neutron-large-ops/b2aecc1/ : SUCCESS in 31m 38s\n- gate-nova-tox-functional http://logs.openstack.org/24/275624/4/check/gate-nova-tox-functional/2ac3571/ : SUCCESS in 17m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/4/check/gate-grenade-dsvm-multinode/ce76965/ : SUCCESS in 1h 00m 53s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ca1300b/ : SUCCESS in 40m 01s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-nova-v20-api/e7ab9b6/ : SUCCESS in 51m 17s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-nova-v20-api-legacy/07c7c26/ : SUCCESS in 57m 47s\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-cells/978218f/ : SUCCESS in 59m 57s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-full-ceph/2e8c45e/ : SUCCESS in 47m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/a0ad737/ : SUCCESS in 47m 00s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/4/check/gate-tempest-dsvm-multinode-full/aec8778/ : FAILURE in 1h 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4062539bf5e20a2813a0cdbce286c33886f8c268","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 19:58:46.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/4/ : FAILURE in 3h 02m 33s","accounts_in_message":[],"_revision_number":4},{"id":"34f0869aafc27a09cc1d7e1ab6ee8c3b29fb35c0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-04 20:44:41.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"491e1279a6c780809494617859e6b60ffe3359af","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 20:47:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/4/check/check-dsvm-tempest-vz7-exe-minimal/b6f189e : SUCCESS in 1h 05m 33s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"2bfc9338f1cc6569f01d74071dcf1c7646110e6f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-05 00:38:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275624/4/Nova-ML2-Sriov : FAILURE in 42m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"2cb3e61cd669426ba4273517bb5cf4843b6689c1","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2016-02-05 03:22:33.000000000","message":"Patch Set 2:\n\nBuild Failed.\n* check-tempest-networking http://intel-openstack-ci-logs.ovh/networking-vif-ci/refs/changes/24/275624/2 : FAILURE\n\nTo re-run the job please post a comment with `intel-networking-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"0e2d8567ec05709748fd5f5b05b8465124cdfeb3","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2016-02-05 07:19:53.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-networking http://intel-openstack-ci-logs.ovh/networking-vif-ci/refs/changes/24/275624/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-networking-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"53d05485f429729104050bea37d346e50999226e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-05 10:51:34.000000000","message":"Patch Set 4:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/4 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":4},{"id":"53929fac9490086dc04626bf71f9073246111f09","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 08:25:43.000000000","message":"Patch Set 2:\n\nBuild Failed.\n* check-tempest-networking http://intel-openstack-ci-logs.ovh/networking-vif-ci/refs/changes/24/275624/2 : FAILURE\n\nTo re-run the job please post a comment with `intel-networking-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"e9df29e0271c39eac4be0b8f7c2b73d0015ebf40","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-10 14:24:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/4/2122 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"2222627ab00c290ced619cc911febd9e34b35209","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 07:29:09.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"ec91c640d5c5cd566574c50aad7beae1a97e4e04","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-03-07 09:48:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d549155321b33b1fb839265642e59a99703c3b74","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 10:38:45.000000000","message":"Patch Set 5:\n\nBuild Failed.\n* check-tempest-networking http://intel-openstack-ci-logs.ovh/networking-vif-ci/refs/changes/24/275624/5 : FAILURE\n\nTo re-run the job please post a comment with `intel-networking-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"63bc8796bf40f1341b3fbf31875a2e44cd86694d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-07 10:50:08.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/5b79fe1286204e35af6222560b4f0eeb/index.html : SUCCESS in 13m 30s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8f81fb77433e48dc97cd998b306c4dcc/index.html : SUCCESS in 18m 15s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/053c179d07474335bbdef8677ebf0d67/index.html : SUCCESS in 12m 57s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/5/check/real-db-upgrade_nova_percona_user_001:th-percona/589157b02a134640bb6d6ea2b406316a/index.html : SUCCESS in 17m 37s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/9d34c6175b1e48a0b0dd9af773fb118d/index.html : SUCCESS in 18m 48s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/5/check/real-db-upgrade_nova_percona_user_002:th-percona/1453f7d1228449aab63d267b349bc10f/index.html : SUCCESS in 18m 22s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"a2929a0626980e323c3ee23e8a2c462f744d2e2f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 10:55:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/24/275624/5/check/gate-nova-pep8/ae19e9e/ : SUCCESS in 3m 12s\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/5/check/gate-nova-docs/4441434//doc/build/html/ : SUCCESS in 4m 07s\n- gate-nova-python27 http://logs.openstack.org/24/275624/5/check/gate-nova-python27/9d401bb/ : SUCCESS in 8m 09s\n- gate-nova-python34 http://logs.openstack.org/24/275624/5/check/gate-nova-python34/549c543/ : SUCCESS in 35m 03s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-full/9600c9f/ : SUCCESS in 42m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-postgres-full/d5e4bbb/ : SUCCESS in 41m 49s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-neutron-full/037670f/ : SUCCESS in 55m 46s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/5/check/gate-grenade-dsvm/e4a71a9/ : SUCCESS in 51m 48s\n- gate-nova-tox-functional http://logs.openstack.org/24/275624/5/check/gate-nova-tox-functional/e2d64a1/ : SUCCESS in 10m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/5/check/gate-grenade-dsvm-multinode/4ef6217/ : SUCCESS in 43m 48s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a45c49b/ : SUCCESS in 40m 36s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-cells/a5b44b4/ : SUCCESS in 29m 00s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-full-ceph/dd989c1/ : SUCCESS in 38m 34s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/29377c8/ : SUCCESS in 38m 32s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/5/check/gate-tempest-dsvm-multinode-full/af5c29f/ : FAILURE in 57m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"301aba9aa80979734246d1e713586ed85e127176","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-07 10:57:57.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/5/check/check-ibm-tempest-dsvm-full/3a6af12/ : SUCCESS in 1h 08m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a2573c0b4664db6133b682f0d27eec5aa16ff08b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-07 11:07:56.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/275624/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"aab816c4f822a8e760672c07cf64f29a8feb28ad","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-07 11:08:51.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/5/8441/index.html : SUCCESS in 01 h 16 m 40 s","accounts_in_message":[],"_revision_number":5},{"id":"162612c5eefa5621e18465c7d52afcd4943270f1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-07 12:15:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/5/check/dsvm-tempest-xen/d7e2196 : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":5},{"id":"c8ed1dc9977197c5c08ca55abe9e30f99934d440","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 13:04:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/5/ : SUCCESS in 2h 00m 15s","accounts_in_message":[],"_revision_number":5},{"id":"bdbe7d7636f2d2ba234eb27ba6df11716c393869","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-07 15:52:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/5/check/check-dsvm-tempest-vz7-exe-minimal/c81109f : SUCCESS in 1h 06m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"2475218b68da4e5dec12876393cbfe36d7ca1255","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-11 14:03:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/5/2832 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"9d45dff00e1186e762673e7b70b7a056ec475ea9","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-20 13:34:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d1aca22c791e05e30e185228cbb677cb9b28d8ba","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-03-20 13:34:52.000000000","message":"Patch Set 6:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/6 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":6},{"id":"8887a432545ed44daa319c432284fbd98bccc070","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-20 13:51:33.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/275624/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"a0f3b7898791fef37f817ee938fe89476edbdd1c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-20 13:53:32.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/bf50a8ee07814e52ba4212831dac3117/index.html : SUCCESS in 16m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f3ecfc2c0cee446f92d63bf214d5936f/index.html : SUCCESS in 17m 31s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6cebf7740e7647919cc863a36b5c5f53/index.html : SUCCESS in 15m 29s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/6/check/real-db-upgrade_nova_percona_user_001:th-percona/19c648c70b6e469b83dc98c0a39d3382/index.html : SUCCESS in 16m 33s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d303364a9d07460684f698421c302d5e/index.html : SUCCESS in 12m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/6/check/real-db-upgrade_nova_percona_user_002:th-percona/38eb0982863d4aedb48480e0ec22d31e/index.html : SUCCESS in 15m 44s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"45da8f1e4e1f54ff005370714b50828d674f092c","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-20 14:09:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c581db86a548adffafdaa07ccd3014c2387cb5bc","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2016-03-20 14:09:32.000000000","message":"Patch Set 7:\n\nBuild Failed.\n* check-tempest-networking http://intel-openstack-ci-logs.ovh/networking-vif-ci/refs/changes/24/275624/7 : FAILURE\n\nTo re-run the job please post a comment with `intel-networking-ci recheck`","accounts_in_message":[],"_revision_number":7},{"id":"b54b7ae0a5cc3be6cbb1320f96d16d7968d9f27e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-20 14:26:21.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/275624/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"1cd0df93fdf07d78ed975da9979a8f06fcc831f1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-20 14:30:36.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/02a2809d52cc47c4b5c5c0bd6d94cf2e/index.html : SUCCESS in 15m 57s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9d3ffb386bf74f93a6401aed34b8577c/index.html : SUCCESS in 16m 53s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1503da60a77b4d59be23ef3a1449dc93/index.html : SUCCESS in 15m 33s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/7/check/real-db-upgrade_nova_percona_user_001:th-percona/74788452b88142d8917fcd0c79986bcc/index.html : SUCCESS in 19m 50s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/b81eeb9ec0314731856e0d675c5d18b0/index.html : SUCCESS in 16m 34s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/7/check/real-db-upgrade_nova_percona_user_002:th-percona/e3fc5010ac76493ca1f3f64aef6d82a9/index.html : SUCCESS in 13m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"2991853ebb229966e15d5b48b0f195323ff36226","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-20 14:30:38.000000000","message":"Topic ovs_offload6 removed","accounts_in_message":[],"_revision_number":7},{"id":"47b8f3cbd81c2ad3c5d4817c4c1e6f019a1ae669","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-20 14:37:59.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/7/ : FAILURE in 28m 10s","accounts_in_message":[],"_revision_number":7},{"id":"5196086e90ecc75c5c305f5daf69b82847aaa423","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-03-20 14:57:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275624/7/Nova-ML2-Sriov : FAILURE in 43m 19s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"71311231c31c49cd11ac710dba1fbb87a0e0e8b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-20 15:25:52.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/7/check/check-ibm-tempest-dsvm-full/8e48c4e/ : FAILURE in 1h 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d9c0f8d78780b8ee319125ba073cce61ea19cfdc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-20 15:27:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/7/check/gate-nova-docs/4a18271//doc/build/html/ : SUCCESS in 5m 08s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/7/check/gate-nova-pep8/2302979/ : SUCCESS in 4m 20s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/7/check/gate-nova-python27-db/3235d8d/ : SUCCESS in 11m 55s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/7/check/gate-nova-python34-db/f7d28b6/ : SUCCESS in 17m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-full/9be2376/ : SUCCESS in 42m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-postgres-full/7b2d05e/ : SUCCESS in 59m 53s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-neutron-full/4cc875c/ : SUCCESS in 1h 13m 56s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/7/check/gate-grenade-dsvm/39fc29c/ : SUCCESS in 52m 39s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/7/check/gate-nova-tox-db-functional/a95e659/ : SUCCESS in 15m 14s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/7/check/gate-grenade-dsvm-multinode/c6637e2/ : SUCCESS in 47m 19s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a9baffa/ : SUCCESS in 46m 38s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-cells/d25b6b4/ : SUCCESS in 35m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/3faeb3a/ : SUCCESS in 53m 21s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/7/check/gate-tempest-dsvm-multinode-full/53cc8b4/ : SUCCESS in 55m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"03fea940186d7991b6da0fbcc13f433f2f8dc723","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-20 15:29:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/7/check/check-dsvm-tempest-vz7-exe-minimal/27b9f50 : SUCCESS in 1h 19m 07s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"6e9de5ecacd97cf6fa6599cb8697445c0888c4c4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-20 15:39:11.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/7/9610/index.html : SUCCESS in 01 h 27 m 41 s","accounts_in_message":[],"_revision_number":7},{"id":"f99b5c5021e218c0c37cafbafbd19756f0ef36cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-03-20 16:00:38.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"d5b0884fa12e20b29f4fbd28969f2cc5ac820ac0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-20 16:28:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/7/3046 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"fa6df23723ca4a5917ca8195dac092b94700a645","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 08:50:13.000000000","message":"Patch Set 7:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/7 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":7},{"id":"7f53f2679dc9debfcdde975e4fa5edbe5a4f6bfd","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-21 09:41:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9281808689b7f7b2ef74ff98c4aaefe46fb0749b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 09:47:18.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- http://10.9.1.52:8080/job/nova-dvsm-full-tempest/15387/ : ABORTED in 5m 24s","accounts_in_message":[],"_revision_number":8},{"id":"63a7fe63d28c6221290c3fe5df02327bdff8a066","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-21 10:02:47.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/275624/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"19f37f8b31833404105c9f49ed7906f0977a4bf9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-21 10:43:54.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/8/check/check-ibm-tempest-dsvm-full/82fa5e5/ : SUCCESS in 1h 02m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6f9842ebe6cf561fdb1705c16f4a9f7478d43cb9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-21 10:57:03.000000000","message":"Patch Set 8: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/8/9649/index.html : SUCCESS in 01 h 13 m 40 s","accounts_in_message":[],"_revision_number":8},{"id":"db0a979eaca7a4db82ad25e437c5c12b412fef64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-21 11:05:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/8/check/gate-nova-docs/6ae1349//doc/build/html/ : SUCCESS in 4m 48s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/8/check/gate-nova-pep8/6b990aa/ : SUCCESS in 2m 34s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/8/check/gate-nova-python27-db/f002d4d/ : SUCCESS in 8m 08s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/8/check/gate-nova-python34-db/bbb7be1/ : SUCCESS in 16m 28s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-full/39ff774/ : SUCCESS in 35m 56s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-postgres-full/e639930/ : SUCCESS in 32m 38s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-neutron-full/d933580/ : SUCCESS in 1h 00m 36s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/8/check/gate-grenade-dsvm/a2d463f/ : SUCCESS in 43m 45s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/8/check/gate-nova-tox-db-functional/8ee2915/ : SUCCESS in 13m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/8/check/gate-grenade-dsvm-multinode/7f986a9/ : SUCCESS in 45m 26s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c134e25/ : SUCCESS in 39m 59s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-cells/d24de17/ : SUCCESS in 35m 39s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/ac5899e/ : SUCCESS in 46m 38s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/8/check/gate-tempest-dsvm-multinode-full/51fda7b/ : SUCCESS in 53m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"271e96f5c992a8db4b2894912c2dc50c65620fc3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 11:33:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/8/check/check-dsvm-tempest-vz7-exe-minimal/4944656 : SUCCESS in 1h 15m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"896f40f182d6583ef382026e377d6d707418fac4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-03-21 11:41:38.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"51d3b7efd38dfed835a0e408babd0fe893d655f1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 11:49:29.000000000","message":"Patch Set 8:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/8 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":8},{"id":"eb2e764c21fdc2a695d3c288d9894cbc2f17f0aa","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-21 13:21:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"49d70be78633366bef0f52660dce8246ffc454ee","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-21 13:38:52.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/69f4b0d8cb014c20a19483bd8c8eaa0d/index.html : SUCCESS in 16m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/2a87dfea325541d6afa3e3088fa5bd68/index.html : SUCCESS in 14m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3c8a5b49223340a4945c10420fa11705/index.html : SUCCESS in 15m 48s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/9/check/real-db-upgrade_nova_percona_user_001:th-percona/d93b0e601ec54c3fadc905bc26f23d19/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/9/check/real-db-upgrade_nova_mysql_user_002:th-mysql/b07eab199ce848328610a5c833d6d58c/index.html : SUCCESS in 15m 44s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/9/check/real-db-upgrade_nova_percona_user_002:th-percona/2dd607d939d54c15b2f0da7cf4d8e97b/index.html : SUCCESS in 15m 00s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"a9d8ee5369c33954b695041d46b6697201504cdf","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-21 13:58:48.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/275624/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"33e4712cb2a74aa4849e8c778dc1f1425bf9f81a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-21 14:25:39.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/9/check/check-ibm-tempest-dsvm-full/861b7ea/ : SUCCESS in 1h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dd2d89536277794068d77007d5645f9c38ef77bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-21 14:40:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/9/check/gate-nova-docs/0f87c98//doc/build/html/ : SUCCESS in 22m 51s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/9/check/gate-nova-pep8/cbc13ac/ : SUCCESS in 3m 14s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/9/check/gate-nova-python27-db/f376710/ : SUCCESS in 12m 10s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/9/check/gate-nova-python34-db/cbb7af7/ : SUCCESS in 15m 45s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-full/93a4afc/ : SUCCESS in 56m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-postgres-full/611a145/ : SUCCESS in 35m 04s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-neutron-full/df5a023/ : SUCCESS in 1h 02m 53s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/9/check/gate-grenade-dsvm/f5aa0a0/ : SUCCESS in 42m 18s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/9/check/gate-nova-tox-db-functional/ccf31b2/ : SUCCESS in 13m 15s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/9/check/gate-grenade-dsvm-multinode/b27245a/ : FAILURE in 27m 59s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3ed9631/ : SUCCESS in 38m 29s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-cells/6d62da3/ : SUCCESS in 38m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/78a7942/ : SUCCESS in 49m 14s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-multinode-full/4cfeaa0/ : FAILURE in 1h 02m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"432932556456f1230179eb98455452fc7611cbef","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-21 14:46:11.000000000","message":"Patch Set 9: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/9/9675/index.html : SUCCESS in 01 h 14 m 59 s","accounts_in_message":[],"_revision_number":9},{"id":"693b98a8ad56206e0c55b6850416f4b035aa3eeb","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 15:14:11.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/9/ : SUCCESS in 1h 52m 23s","accounts_in_message":[],"_revision_number":9},{"id":"feb263288fd4d5554cd33e9d7d41fcf38e227cf7","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-03-21 16:27:36.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5cade02c9203234240c50cd289d1e18ce7420ee1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-21 16:35:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/9/check/dsvm-tempest-xen/ba4a760 : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":9},{"id":"7ee29ef4704d77cc36d7f47d3bd5b316d4a83aca","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 16:53:49.000000000","message":"Patch Set 9:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/9 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":9},{"id":"e069e7f1dafd7c8dacd1ec3c782f431f6cbedcf1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 17:01:48.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/9/check/check-dsvm-tempest-vz7-exe-minimal/04f0d88 : SUCCESS in 1h 14m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"c4cdd395933a57ab568ab7c23ee3cb8dc06ed55c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 18:02:23.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/9/ : SUCCESS in 1h 34m 24s","accounts_in_message":[],"_revision_number":9},{"id":"a54bfee6f54b9b426c8d3f9da993b61c2704baa3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-21 19:11:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/9/check/gate-nova-docs/d274fa9//doc/build/html/ : SUCCESS in 6m 13s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/9/check/gate-nova-pep8/bfacef4/ : SUCCESS in 4m 41s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/9/check/gate-nova-python27-db/bee7493/ : SUCCESS in 10m 50s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/9/check/gate-nova-python34-db/e2c3c12/ : SUCCESS in 18m 19s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-full/c9739a2/ : SUCCESS in 2h 40m 03s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-postgres-full/6653b17/ : SUCCESS in 35m 24s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-neutron-full/f4fcdee/ : SUCCESS in 54m 10s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/9/check/gate-grenade-dsvm/e68be29/ : SUCCESS in 35m 55s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/9/check/gate-nova-tox-db-functional/5c4954f/ : SUCCESS in 14m 24s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/9/check/gate-grenade-dsvm-multinode/4bcfec4/ : SUCCESS in 45m 24s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/cadf28a/ : SUCCESS in 22m 55s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-cells/ced3707/ : SUCCESS in 43m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/8ffa3ec/ : SUCCESS in 42m 29s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/9/check/gate-tempest-dsvm-multinode-full/43329cd/ : SUCCESS in 1h 07m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"63bf5271a2c0c562dfbc126cba73f4913185cb5a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-03-21 19:54:04.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"a74d325aa6a2c0709b103f703db5ff08a7e23d93","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-03-21 23:45:30.000000000","message":"Patch Set 9:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/9 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":9},{"id":"466842baca5f43c579191ad6951c1e952be0caed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-23 04:02:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/9/3084 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"7ac307a1a67f8cb0ff35684c5e8d847fd137a4ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-29 21:06:13.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"877d73a8b08bd0fe18a2c91e7be0bacf2d2ab64b","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-03-30 07:30:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6a388f35084d20d6c6b32f133f4bd91274bba14a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-30 07:47:55.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/275624/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"a8cb8139af5bb069cbc43e9e29e1f31677daa508","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 08:27:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/10/check/gate-nova-docs/30df290//doc/build/html/ : SUCCESS in 5m 21s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/10/check/gate-nova-pep8/1cda7c5/ : SUCCESS in 2m 40s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/10/check/gate-nova-python27-db/5ee6350/ : SUCCESS in 8m 04s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/10/check/gate-nova-python34-db/2f0717c/ : SUCCESS in 11m 44s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-full/a6e9174/ : SUCCESS in 48m 53s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-postgres-full/9abe2bc/ : SUCCESS in 46m 36s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-neutron-full/3f60799/ : SUCCESS in 53m 56s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/10/check/gate-grenade-dsvm/01ff6e0/ : SUCCESS in 50m 19s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/10/check/gate-nova-tox-db-functional/5a6756b/ : SUCCESS in 10m 22s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/10/check/gate-grenade-dsvm-multinode/0bf6684/ : SUCCESS in 29m 41s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f8d4fd1/ : SUCCESS in 35m 00s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-cells/a5413c4/ : SUCCESS in 27m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/f9261ab/ : SUCCESS in 37m 38s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-multinode-full/3880049/ : SUCCESS in 50m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e124ec4c3495b4111a9784b2eae571c8496a812f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-30 08:31:04.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/10/check/check-ibm-tempest-dsvm-full/0ab082d/ : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7445798646814cf2d42b06dc3203d0f6cc8edb3a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-30 08:41:27.000000000","message":"Patch Set 10: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/10/10581/index.html : SUCCESS in 01 h 08 m 40 s","accounts_in_message":[],"_revision_number":10},{"id":"a52be6d675ce2d5d438f5ea529e47afb745bd76e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-30 08:55:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/10/check/dsvm-tempest-xen/788aec9 : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":10},{"id":"b9a26cedd240286c01a7dac21e28c91d467199e8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-30 09:02:37.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/10/ : SUCCESS in 1h 31m 37s","accounts_in_message":[],"_revision_number":10},{"id":"ddb5eb22c34587d00b52ebe2ce906975e1a903ae","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-30 12:48:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/10/check/check-dsvm-tempest-vz7-exe-minimal/1603b43 : SUCCESS in 1h 16m 30s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"138cbf02093d3d5cb1624eece300b56e80f13e5e","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-03-30 13:14:25.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/24/275624/10/check/dsvm-tempest-kvm/64f7940 : SUCCESS in 1h 17m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"310bea9bdf654c32abb31ba19ac0a17fc5f8f700","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-03-30 19:34:04.000000000","message":"Patch Set 10:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/24/275624/10 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":10},{"id":"b0380d6f72b1f30817d2d335f6a04f89f8bbb469","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-31 06:10:31.000000000","message":"Patch Set 10:\n\nrecheck nova-mlnx","accounts_in_message":[],"_revision_number":10},{"id":"53086722a98af8ec3055283c965d0feefbdf4121","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-31 07:22:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/10/check/gate-nova-docs/357dd71//doc/build/html/ : SUCCESS in 8m 45s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/10/check/gate-nova-pep8/0fdc277/ : SUCCESS in 4m 14s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/10/check/gate-nova-python27-db/d0ebbe4/ : SUCCESS in 10m 23s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/10/check/gate-nova-python34-db/e328aac/ : SUCCESS in 17m 32s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-full/d5a356e/ : SUCCESS in 1h 03m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-postgres-full/bc4c8c3/ : SUCCESS in 43m 33s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-neutron-full/bc82b3e/ : SUCCESS in 1h 09m 33s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/10/check/gate-grenade-dsvm/9e9c68f/ : SUCCESS in 49m 05s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/10/check/gate-nova-tox-db-functional/a6c655c/ : SUCCESS in 11m 55s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/10/check/gate-grenade-dsvm-multinode/0d0cff0/ : SUCCESS in 49m 07s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/1b04e8d/ : SUCCESS in 23m 17s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-cells/ed1ee9c/ : SUCCESS in 40m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/6cf749c/ : SUCCESS in 57m 59s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-multinode-full/ce5a919/ : SUCCESS in 53m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cb453fdeec01be57fcbe2fcc3d1d530efc8a60e3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-31 07:44:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/10/ : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":10},{"id":"b3e8be87df5e636596e6c8eb0b65aeec23df888b","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-03-31 09:28:38.000000000","message":"Patch Set 10:\n\nrecheck nova-mlnx","accounts_in_message":[],"_revision_number":10},{"id":"77131c0ce0861407c5cf48119a5d707c1dca421e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-31 11:03:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/10/check/gate-nova-docs/4cc7fa9//doc/build/html/ : SUCCESS in 8m 08s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/10/check/gate-nova-pep8/0c6113d/ : SUCCESS in 3m 57s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/10/check/gate-nova-python27-db/6715961/ : SUCCESS in 9m 27s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/10/check/gate-nova-python34-db/7ba4412/ : SUCCESS in 15m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-full/4261959/ : FAILURE in 20m 48s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-postgres-full/e2bdd8b/ : SUCCESS in 1h 09m 17s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-neutron-full/815b384/ : SUCCESS in 1h 33m 11s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/10/check/gate-grenade-dsvm/e605c24/ : SUCCESS in 27m 20s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/10/check/gate-nova-tox-db-functional/ba5a520/ : SUCCESS in 11m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/10/check/gate-grenade-dsvm-multinode/fea09f5/ : SUCCESS in 50m 30s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/7eb211c/ : SUCCESS in 33m 46s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-cells/1bfec01/ : SUCCESS in 40m 20s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/7e3b7a7/ : SUCCESS in 43m 44s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/10/check/gate-tempest-dsvm-multinode-full/f5d718a/ : FAILURE in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"50c2254eeeb58f59a88451efc55f46cd96629f20","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-03-31 11:03:26.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1298006 https://bugs.launchpad.net/bugs/1282876\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"a49d5f56972baa6b76ae70447e23b24245681123","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-31 11:03:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/10/ : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":10},{"id":"6911a86d19edad57a75c8d1e1019cc991a74bd92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-31 13:19:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/10/3430 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"e52960091960c8aa1792095200a4d49f2b1ba616","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-03-31 18:45:49.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"db4cbf012e04b528726d2fd5a9654e100d837588","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-02 14:11:44.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d3a576450f454095a53050b71fb15e82/index.html : SUCCESS in 14m 35s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d8be788a502e4003851ea3f17cbd8a08/index.html : SUCCESS in 17m 03s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/81e9feba725e4aa7ad5f2fc43498579c/index.html : SUCCESS in 16m 11s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/10/check/real-db-upgrade_nova_percona_user_001:th-percona/25796e3cd29c428ea24a3fca6736ae90/index.html : SUCCESS in 14m 33s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"a4f933d6cecf45f65ce33c64b9537e3d16b006b1","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2016-04-04 15:53:26.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"bb9e7ffed0a5a7dd9eebb462c9588e8b4c3cda6b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-04 16:54:05.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/11/check/check-ibm-tempest-dsvm-full/42048b0/ : SUCCESS in 1h 00m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8c43d301ff7766d3ea745b7b0e807a77fc26b6bc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-04 17:16:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/11/check/dsvm-tempest-xen/dc1ac71 : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":11},{"id":"205ee57df7367162b76296850a4cf5b39fdc2fba","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-04 17:19:23.000000000","message":"Patch Set 11: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/11/11091/index.html : SUCCESS in 01 h 22 m 49 s","accounts_in_message":[],"_revision_number":11},{"id":"3b426a4ae562a9f4945920233b36b158b5d7ca7b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-04 17:28:20.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/11/ : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":11},{"id":"669bbe2e63fd77128c885097cba4ac58231d6a07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-04 18:18:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/11/check/gate-nova-docs/aa72425//doc/build/html/ : SUCCESS in 2m 29s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/11/check/gate-nova-pep8/31fdcd8/ : SUCCESS in 4m 47s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/11/check/gate-nova-python27-db/038db00/ : SUCCESS in 13m 05s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/11/check/gate-nova-python34-db/0b7e110/ : SUCCESS in 14m 07s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-full/ee36f9f/ : SUCCESS in 38m 01s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-postgres-full/97f12a5/ : SUCCESS in 46m 29s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-neutron-full/5616a98/ : SUCCESS in 57m 50s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/11/check/gate-grenade-dsvm/922fd95/ : SUCCESS in 27m 12s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/11/check/gate-nova-tox-db-functional/f4ca220/ : SUCCESS in 12m 43s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/11/check/gate-grenade-dsvm-multinode/e0b9952/ : SUCCESS in 46m 36s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/51985d3/ : SUCCESS in 24m 48s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-cells/9d32a7c/ : SUCCESS in 29m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/f6039ef/ : SUCCESS in 37m 15s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/11/check/gate-tempest-dsvm-multinode-full/66711c4/ : FAILURE in 52m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"468b21df7610d2fbf221d6ca1ee4e65f2c0729f6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-04 19:16:35.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"a742b885a52be7f0d7ff3fb1cad4983560e5c714","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-04 19:32:10.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/pci/275624/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"c2668cc5e213429478025ec857a7b1d71a97db9c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-04-04 21:04:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/24/275624/11/check-nova/Nova-ML2-Sriov/aa7948b : SUCCESS in 48m 12s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/24/275624/11/check-nova/Nova-MACVTAP-ML2-Sriov/746d93c : SUCCESS in 44m 18s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"34df64d8938095c49fa89a312e696404cc20b64d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-04-04 22:16:02.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/24/275624/11/check/dsvm-tempest-kvm/7c02610 : SUCCESS in 37m 59s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"643002c87f32acd7b475a733770001a4907c9031","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-04 22:52:33.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/11/check/check-dsvm-tempest-vz7-exe-minimal/7443e78 : SUCCESS in 1h 22m 34s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"494319d4d7f839aa405426c9cef924ee5ad9cefd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-05 12:05:55.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/00b70d08c4bb4735abab2b5ac9980d86/index.html : SUCCESS in 15m 02s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6839f5b779494c368161996bd3ac66c5/index.html : SUCCESS in 12m 22s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a8ce2d6f64074173bb6ece1071843a4e/index.html : SUCCESS in 13m 32s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/11/check/real-db-upgrade_nova_percona_user_001:th-percona/da79d0b7b92f444b80650b63cfbc33b3/index.html : SUCCESS in 13m 06s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"f8eec502754d32e5b53a4285188f9817f0849e61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-06 06:41:33.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/11/3618 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"7880bbbf3fef147439a7bbd0533a84b253077941","author":{"_account_id":19726,"name":"Szilard Cserey","email":"szilard.cserey@gmail.com","username":"szilard"},"date":"2016-04-16 00:07:10.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"fc2a2f884d0b5cf2b488d22a7a22087f6a423366","author":{"_account_id":19726,"name":"Szilard Cserey","email":"szilard.cserey@gmail.com","username":"szilard"},"date":"2016-04-17 23:27:10.000000000","message":"Removed the following votes:\n\n* Code-Review+1 by Szilard Cserey \u003cszilard.cserey@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":11},{"id":"b63b6f4bfdb7424c2222c9c19437f7698f5b6179","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-05-23 16:38:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1f2eb8de5b16dadc05d1878f62570fb00b2f1672","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-23 18:01:27.000000000","message":"Patch Set 12: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/12/16105/index.html : SUCCESS in 01 h 18 m 33 s","accounts_in_message":[],"_revision_number":12},{"id":"834cc6a3ca71b713fb589ccb6c42a2ed6acd3627","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-23 18:36:56.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/12/check/gate-nova-docs/21cbc75//doc/build/html/ : SUCCESS in 4m 50s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/12/check/gate-nova-pep8/b80565e/ : FAILURE in 5m 33s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/12/check/gate-nova-python27-db/26bbd6a/ : SUCCESS in 13m 36s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/12/check/gate-nova-python34-db/380f990/ : SUCCESS in 9m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-full/7e5ca6b/ : SUCCESS in 38m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-postgres-full/47d2caa/ : SUCCESS in 41m 08s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-neutron-full/6458058/ : SUCCESS in 1h 01m 30s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/12/check/gate-grenade-dsvm/cbc20dd/ : SUCCESS in 40m 28s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/12/check/gate-nova-tox-db-functional/d6085af/ : SUCCESS in 11m 48s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/12/check/gate-grenade-dsvm-multinode/a4b57a1/ : SUCCESS in 47m 30s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c8a9ef8/ : SUCCESS in 43m 34s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-cells/f9defde/ : SUCCESS in 36m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-full-devstack-plugin-ceph/3760171/ : SUCCESS in 38m 07s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-multinode-full/cbebc08/ : FAILURE in 56m 30s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/24/275624/12/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/887a97a/ : SUCCESS in 1h 04m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"35b7172c47e2c96929cfbe8a6c4f87cb6563a0fd","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-23 18:54:04.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/275624/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"99373601ed5c8291de457720a84ff16b5215a79b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-23 20:04:23.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/12/check/dsvm-tempest-xen/d1253ab : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":12},{"id":"4cb4b617379467a1052228998c84e2034c343917","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-23 20:16:52.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/12/ : SUCCESS in 2h 38m 58s","accounts_in_message":[],"_revision_number":12},{"id":"1fce5a01bcfc6530e57d4156a72a3ccc5ff6ab04","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-23 22:06:57.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8a3904398d4e4d18a609f5a55663b1af/index.html : SUCCESS in 14m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/02ac00659a714fe3a9954afe9cd46ffa/index.html : SUCCESS in 13m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/68f728dca47d43dbaafe3814fd66c183/index.html : SUCCESS in 14m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/12/check/real-db-upgrade_nova_percona_user_001:th-percona/8bf217cf0923445db8134f0350d27a08/index.html : SUCCESS in 14m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"1e73e713ac0ca44c48d974da8c7d735c908cf55f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 01:18:17.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/12/check/check-dsvm-tempest-vz7-exe-minimal/94dd289 : SUCCESS in 2h 19m 42s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"c240100668807756c733fea483ab0513ecf7d063","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-05-24 09:34:04.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d834fe1da4af40e1618ac4ebd3e523e1f443aafe","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-24 10:03:14.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/275624/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"4e050e3ee364f7991a0186f7df9955a8c06d0d7d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-24 10:47:00.000000000","message":"Patch Set 13: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/13/16178/index.html : SUCCESS in 01 h 10 m 09 s","accounts_in_message":[],"_revision_number":13},{"id":"cbdd8a6e65723993b44777f06c011a32ca73db09","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-24 11:06:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/13/check/dsvm-tempest-xen/71337c1 : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":13},{"id":"3ab5c93f12eeaf4cebf09487f18317fe2c88c12d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-24 11:10:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/13/check/gate-nova-docs/bec0cd6//doc/build/html/ : SUCCESS in 4m 04s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/13/check/gate-nova-pep8/8c3a556/ : SUCCESS in 6m 07s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/13/check/gate-nova-python27-db/bb4e53b/ : SUCCESS in 9m 58s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/13/check/gate-nova-python34-db/1e3423a/ : SUCCESS in 9m 47s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-full/a31531c/ : SUCCESS in 46m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-postgres-full/32613e1/ : SUCCESS in 51m 24s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-neutron-full/42ecad2/ : SUCCESS in 59m 46s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/13/check/gate-grenade-dsvm/6d38ec4/ : SUCCESS in 43m 26s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/13/check/gate-nova-tox-db-functional/b8be5ac/ : SUCCESS in 11m 27s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/13/check/gate-grenade-dsvm-multinode/0ac5734/ : SUCCESS in 48m 39s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9628216/ : SUCCESS in 38m 33s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-cells/e5585db/ : SUCCESS in 39m 54s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph/75fca12/ : SUCCESS in 38m 04s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-multinode-full/bd326c9/ : SUCCESS in 1h 28m 44s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/24/275624/13/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/e042539/ : SUCCESS in 1h 02m 11s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f5bdf3be37909d07efff7acff1ddf6e3e1c2b9f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-24 11:16:20.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/13/check/check-ibm-tempest-dsvm-full/1a456e7/ : SUCCESS in 1h 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ed0914e70628cf671036c9bea6c4b1a4babea003","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 11:44:01.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/13/ : SUCCESS in 2h 09m 27s","accounts_in_message":[],"_revision_number":13},{"id":"3a31cdc3087a842edec1757576b955cb5f2eb50d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-05-24 12:31:35.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-13 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":13},{"id":"b63bfe6e28e7b10bdb49976dd14b93f3a55af023","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 12:42:28.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/13/check/check-dsvm-tempest-vz7-exe-minimal/f604c21 : SUCCESS in 1h 06m 08s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"308ed980bb5a0d324c9ce39cac968477ca71d0f4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 14:36:50.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/24/275624/13/check/dsvm-tempest-kvm/17b259d : SUCCESS in 1h 08m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"aee523b66a2420de91543cab1c61fd4283090fdc","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-25 03:30:26.000000000","message":"Patch Set 13: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/13/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f8576221a8a847d9a2e3786b6c45dd92/index.html : SUCCESS in 13m 51s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/13/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9647d12d9fb3452c87b67d44e48e5182/index.html : SUCCESS in 12m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/13/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c8f80fb4a95d405a82963b3b48bb3103/index.html : SUCCESS in 16m 11s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/13/check/real-db-upgrade_nova_percona_user_001:th-percona/83e78b02dbcf4d9097f4b96ec7b409c4/index.html : SUCCESS in 15m 33s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":13},{"id":"b35d6ebe01187c2da14fd8a7d06a0199189965db","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-25 04:58:59.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/24/275624/13/check/tempest-dsvm-full-nfv/8121daa : FAILURE in 1h 38m 59s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/24/275624/13/check/tempest-dsvm-intel-nfv/ce5862c : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":13},{"id":"47620b2b14bd581beef89002901d09f126f51f4e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-25 10:31:54.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/24/275624/12/check/tempest-dsvm-full-nfv/d9988f9 : FAILURE in 1h 41m 40s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/24/275624/12/check/tempest-dsvm-intel-nfv/c4cb7fe : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":12},{"id":"6cf81d12b4057b57aa0dbb800e7c6119f1219689","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-05-26 04:29:21.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275624/13/4843 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"cb3081200eb01987c396cb35247deda605e15aad","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-05-30 09:39:52.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7c21654ca5cb5e3f8de4c7f20cc4566059e492c0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-30 10:33:35.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/275624/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"358a5313b66ff7066a87cd9c29fc4d6a961f698c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-30 10:44:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/14/check/gate-nova-docs/b3290b2//doc/build/html/ : SUCCESS in 4m 34s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/14/check/gate-nova-pep8/6c7fc06/ : SUCCESS in 5m 36s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/14/check/gate-nova-python27-db/dd24a9e/ : FAILURE in 12m 06s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/14/check/gate-nova-python34-db/ea49853/ : FAILURE in 13m 28s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-full/f6be8d3/ : SUCCESS in 47m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-postgres-full/9fd90c0/ : SUCCESS in 41m 30s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-neutron-full/1da0412/ : SUCCESS in 56m 41s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/14/check/gate-grenade-dsvm/ced529b/ : SUCCESS in 48m 47s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/14/check/gate-nova-tox-db-functional/285df2e/ : FAILURE in 9m 27s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/14/check/gate-grenade-dsvm-multinode/66b1370/ : SUCCESS in 44m 24s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/46fbc46/ : FAILURE in 26m 39s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-cells/c8a77cb/ : SUCCESS in 28m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph/25f2743/ : SUCCESS in 44m 17s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-multinode-full/c91f578/ : FAILURE in 1h 03m 52s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/80259a0/ : SUCCESS in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"abe236403a8abc88634580c3ec3ba8fd7fe3e524","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-30 10:56:05.000000000","message":"Patch Set 14: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/14/16859/index.html : SUCCESS in 01 h 12 m 24 s","accounts_in_message":[],"_revision_number":14},{"id":"c35a7448d9378a7ca71c0b88a12a0c01dd877349","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-30 10:58:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/14/check/dsvm-tempest-xen/bd45e5d : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":14},{"id":"3344bb7285eec965dea0dc09075114441628ca98","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-30 11:31:27.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/14/check/check-ibm-tempest-dsvm-full/a7eda62/ : SUCCESS in 1h 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"558f8386744278ac3b407c317995a5d320d887dd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-30 12:41:10.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/24/275624/14/check-nova/Nova-ML2-Sriov/c358aee : SUCCESS in 33m 54s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/24/275624/14/check-nova/Nova-MACVTAP-ML2-Sriov/30939c9 : SUCCESS in 34m 42s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":14},{"id":"071fd7ca60ce39304d2997c405f4f023eb6b5105","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-30 12:45:28.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/14/ : SUCCESS in 3h 05m 05s","accounts_in_message":[],"_revision_number":14},{"id":"efbdebd7c50afc58afa6d3daba68c278550ccf66","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-30 14:49:19.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/24/275624/14/check/tempest-dsvm-full-nfv/6c1de22 : SUCCESS in 1h 31m 57s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/24/275624/14/check/tempest-dsvm-intel-nfv/f97e6df : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":14},{"id":"dd1dedecc76efc8b0f902ab2b337a5e5c7407fb7","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-02 10:42:39.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e5e72eea792e48b0b1cf405cbf48192c/index.html : SUCCESS in 21m 07s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/63b6e3bfeb07483680657ef53d9b2ea7/index.html : SUCCESS in 18m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/24/275624/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1dd33c74919d46c4a636ec3091ef521b/index.html : SUCCESS in 24m 29s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/24/275624/14/check/real-db-upgrade_nova_percona_user_001:th-percona/a60cb8ff8fcb444dbb4e43cffd5f4bb4/index.html : SUCCESS in 19m 11s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"51eefd8351088ad27912f3c18b9968f9dd296ab5","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-06-02 11:23:50.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"0dc6014636ba01af480252a67e749f5376020e1b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-02 12:25:29.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/14/check/check-ibm-tempest-dsvm-full/ff44127/ : SUCCESS in 59m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0926bbdeaa5665680c51030d87e758fd7863c2b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-02 12:28:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/24/275624/14/check/gate-nova-docs/2e5d9fd//doc/build/html/ : SUCCESS in 4m 06s\n- gate-nova-pep8 http://logs.openstack.org/24/275624/14/check/gate-nova-pep8/cf4b584/ : SUCCESS in 5m 45s\n- gate-nova-python27-db http://logs.openstack.org/24/275624/14/check/gate-nova-python27-db/801d89e/ : SUCCESS in 8m 30s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/14/check/gate-nova-python34-db/3759652/ : SUCCESS in 10m 32s\n- gate-tempest-dsvm-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-full/8fa1382/ : SUCCESS in 50m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-postgres-full/d8f1deb/ : SUCCESS in 45m 08s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-neutron-full/7ca277a/ : SUCCESS in 52m 17s\n- gate-grenade-dsvm http://logs.openstack.org/24/275624/14/check/gate-grenade-dsvm/e968497/ : SUCCESS in 41m 32s\n- gate-nova-tox-db-functional http://logs.openstack.org/24/275624/14/check/gate-nova-tox-db-functional/a26c178/ : SUCCESS in 10m 58s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/14/check/gate-grenade-dsvm-multinode/31288a6/ : SUCCESS in 45m 24s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/11ec1f1/ : SUCCESS in 38m 48s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-cells/46f7649/ : SUCCESS in 37m 35s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph/174ae3b/ : SUCCESS in 44m 34s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-multinode-full/c9fa0bf/ : FAILURE in 56m 27s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/24/275624/14/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/1cdafb9/ : SUCCESS in 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fe5e0fdb39c8aa283d05b48d789b5b27e90c0961","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-02 12:48:57.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/24/275624/14/check/tempest-dsvm-full-nfv/5152bd9 : SUCCESS in 1h 19m 18s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/24/275624/14/check/tempest-dsvm-intel-nfv/eaeaaec : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":14},{"id":"2ce55cf60abcfda2ce59645ef3414477a7b401b5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-02 15:07:31.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/14/ : FAILURE in 3h 43m 18s","accounts_in_message":[],"_revision_number":14},{"id":"54c9c8179dadf205af0b602d074b21dc603841ad","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-19 16:55:39.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1371911be3116b752d91724710ffe7ba21d06e5d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-19 16:56:18.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5848fd070d20d210131808236044f6ce25aa9334","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-19 17:19:25.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/275624/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"10137e38de172cc9f26fdcc2c1348bb5aa4b5069","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-19 18:23:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/24/275624/15/check/gate-nova-docs-ubuntu-xenial/fdb4038//doc/build/html/ : SUCCESS in 3m 51s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-nova-pep8-ubuntu-xenial/13c0e36/ : SUCCESS in 5m 59s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-nova-python27-db-ubuntu-xenial/406b2d9/ : SUCCESS in 19m 03s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/15/check/gate-nova-python34-db/95d814b/ : SUCCESS in 12m 34s\n- gate-nova-python35-db-nv http://logs.openstack.org/24/275624/15/check/gate-nova-python35-db-nv/1478abc/ : SUCCESS in 12m 17s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/24/275624/15/check/gate-grenade-dsvm-neutron-ubuntu-trusty/30e9f94/ : SUCCESS in 48m 14s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3437cda/ : SUCCESS in 1h 03m 42s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-full-ubuntu-xenial/7bf956b/ : SUCCESS in 51m 30s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/24/275624/15/check/gate-grenade-dsvm-ubuntu-trusty/c3788d3/ : SUCCESS in 27m 38s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-nova-tox-db-functional-ubuntu-xenial/fdbb44a/ : SUCCESS in 17m 49s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/24/275624/15/check/gate-grenade-dsvm-multinode/403081d/ : SUCCESS in 47m 46s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-cells-ubuntu-xenial/60fbd68/ : SUCCESS in 35m 26s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8828edb/ : SUCCESS in 1h 04m 43s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/1293bbe/ : SUCCESS in 51m 14s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/0f6d467/ : SUCCESS in 1h 12m 31s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/b36bfbf/ : SUCCESS in 54m 05s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/995d43f/ : SUCCESS in 30m 41s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/24/275624/15/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/976c2e1/ : SUCCESS in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8f1df10d9043236b69152a4fc0253a36017ebcad","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-19 18:25:33.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/15/check/dsvm-tempest-xen/a586f5a : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":15},{"id":"4f7c2c40b70d4192f2bc8c34e365f98979815065","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-19 18:36:41.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/24/275624/15/check-nova/Nova-ML2-Sriov/4de7bf3 : SUCCESS in 34m 34s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/24/275624/15/check-nova/Nova-MACVTAP-ML2-Sriov/98373d3 : SUCCESS in 33m 32s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"dffd6745951200902dd3ed62501ff2c22f2095f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-19 18:59:39.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"f0ba62dca025bbd9e6833f7b7935eb829c5c6a52","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-19 19:46:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/15/check/dsvm-tempest-nova-network/30b4586 : SUCCESS in 1h 21m 08s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/15/check/dsvm-tempest-neutron-network/64d51d8 : SUCCESS in 1h 55m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b65268840647d10c3468f2d2fd4ded2938ea61b4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-19 20:25:20.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/24/275624/15/check/tempest-dsvm-full-nfv-xenial/b6086d3 : SUCCESS in 1h 48m 28s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/24/275624/15/check/tempest-dsvm-intel-nfv-xenial/51cf63c : SUCCESS in 35m 28s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/24/275624/15/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a0acd69 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":15},{"id":"29baca8540be94c73b76fada2a841b4154f52242","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-19 20:52:54.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/15/ : FAILURE in 3h 56m 11s","accounts_in_message":[],"_revision_number":15},{"id":"ce4f32df3c7f31b1522b6335c01be7dd432682df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-22 21:43:08.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/275624/15/1171/ : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"b58c63c92272169452a5fa1c566646479e0cff4b","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-11-14 11:09:21.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"add49e642f72d68bc1e9b7e8fc7dac49b38945e3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-14 11:30:34.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/pci/275624/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"e4c72b8a0ab41d642c9f2494f6e8261783528b76","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-14 12:06:33.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/24/275624/16/check/gate-nova-docs-ubuntu-xenial/174a18b//doc/build/html/ : SUCCESS in 2m 49s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-nova-pep8-ubuntu-xenial/4594537/ : SUCCESS in 6m 23s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-nova-python27-db-ubuntu-xenial/dee055e/ : SUCCESS in 22m 00s\n- gate-nova-python34-db http://logs.openstack.org/24/275624/16/check/gate-nova-python34-db/86146e3/ : SUCCESS in 11m 03s\n- gate-nova-python35-db-nv http://logs.openstack.org/24/275624/16/check/gate-nova-python35-db-nv/1e9bb88/ : SUCCESS in 7m 56s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/30a9395/ : SUCCESS in 47m 59s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-grenade-dsvm-neutron-ubuntu-xenial/116ba94/ : SUCCESS in 31m 18s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-grenade-dsvm-ubuntu-xenial/3631929/ : SUCCESS in 26m 05s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-nova-tox-db-functional-ubuntu-xenial/776948b/ : SUCCESS in 13m 40s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/19478d0/ : SUCCESS in 38m 40s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/24/275624/16/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/bc288ba/ : SUCCESS in 31m 59s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-cells-ubuntu-xenial/c6e9f66/ : SUCCESS in 33m 14s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b7daee0/ : SUCCESS in 55m 22s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/4290382/ : SUCCESS in 44m 58s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/ed8d14f/ : SUCCESS in 51m 07s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/c0a7c5f/ : SUCCESS in 46m 03s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d580d75/ : SUCCESS in 39m 01s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/24/275624/16/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/de256d0/ : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cef61ca7ad752242e031a9f3a43ad8166b88d0af","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-14 12:08:45.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/24/275624/16/check/check-dsvm-tempest-vz7-exe-minimal/5f8a6d4 : SUCCESS in 58m 55s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"3929e77fc9c7b78d950a93a4a00e3f2f2eced2fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-14 12:21:35.000000000","message":"Patch Set 16:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/16/check/tempest-dsvm-full-xenial/1297fab/ : FAILURE in 1h 11m 45s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/24/275624/16/check/tempest-dsvm-neutron-full-xenial/d318b30/ : SUCCESS in 1h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5ca18aeb4952369857ea70176b1dc70d8bea1d0a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-14 12:23:01.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/24/275624/16/check/dsvm-tempest-kvm/6d28048 : SUCCESS in 1h 13m 05s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"00f4a1931b11d2cdcd7be37cd19484ac6d861d62","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-14 12:45:21.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/24/275624/16/check-nova/Nova-ML2-Sriov/632642e : SUCCESS in 1h 04m 06s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/24/275624/16/check-nova/Nova-MACVTAP-ML2-Sriov/b029a86 : SUCCESS in 1h 04m 43s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":16},{"id":"84c4a5c818dc788ec7abd327c1bcfdc0e4f0c7ac","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-14 12:59:00.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275624/16/ : SUCCESS in 1h 48m 56s","accounts_in_message":[],"_revision_number":16},{"id":"b060685fc91e3e640a83f7bdef52035a651d44a6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-14 13:15:15.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/24/275624/16/check/dsvm-tempest-xen/0590fd0 : SUCCESS in 2h 03m 26s","accounts_in_message":[],"_revision_number":16},{"id":"0a706b1704679be3c50ed16a3d8feeb7f16ab7ab","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-14 13:25:45.000000000","message":"Patch Set 16:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-275624-16 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":16},{"id":"4a24c8bafd5278c280c21421880ef9ed4dcb123e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-14 13:49:40.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/24/275624/16/check/dsvm-tempest-neutron-network/803b7e6 : FAILURE in 2h 31m 43s","accounts_in_message":[],"_revision_number":16},{"id":"4485fbf5fabb954058c735f390c31bbc6c3d2f53","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-14 14:39:44.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/24/275624/16/check/tempest-dsvm-full-nfv-xenial/9531afd : SUCCESS in 2h 35m 22s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/24/275624/16/check/tempest-dsvm-intel-nfv-xenial/151ba87 : SUCCESS in 42m 06s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/24/275624/16/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/6d61f63 : SUCCESS in 1h 38m 21s","accounts_in_message":[],"_revision_number":16},{"id":"3642292170203576d1244decbccc02996597940b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-14 19:26:56.000000000","message":"Patch Set 16: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/275624/16/2035/ : SUCCESS\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/275624/16/2035/ : SUCCESS (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":16},{"id":"61a7764e3d1327c4e61495dae6948e4d09ef3bb9","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-11-16 06:20:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"5897f039c857499d03a4722faebef7aed6521ef9","revisions":{"89424b18931939d50016dd6b4438a15e0cc392f0":{"kind":"REWORK","_number":1,"created":"2016-02-03 11:02:20.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/1"}}},"commit":{"parents":[{"commit":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","subject":"Merge \"remove the wrong param of fake_db_migration initiation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-02-03 11:00:26.000000000","tz":120},"subject":"[WIP] POC OVS offload","message":"[WIP] POC OVS offload\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89424b18931939d50016dd6b4438a15e0cc392f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89424b18931939d50016dd6b4438a15e0cc392f0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3b4f9021e62e6eee265c826adec9d3727fc469ea":{"kind":"REWORK","_number":2,"created":"2016-02-04 12:55:19.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/2"}}},"commit":{"parents":[{"commit":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","subject":"Merge \"remove the wrong param of fake_db_migration initiation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-02-04 10:49:53.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b4f9021e62e6eee265c826adec9d3727fc469ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b4f9021e62e6eee265c826adec9d3727fc469ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6bfd3d5187681501b702a6110ed39a62031aec2d":{"kind":"REWORK","_number":3,"created":"2016-02-04 15:22:39.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/3"}}},"commit":{"parents":[{"commit":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","subject":"Merge \"remove the wrong param of fake_db_migration initiation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-02-04 15:21:59.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6bfd3d5187681501b702a6110ed39a62031aec2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6bfd3d5187681501b702a6110ed39a62031aec2d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5979762cf1994e0a3c3b1c79f76ee8ba89a5547a":{"kind":"REWORK","_number":4,"created":"2016-02-04 16:17:17.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/4"}}},"commit":{"parents":[{"commit":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","subject":"Merge \"remove the wrong param of fake_db_migration initiation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-02-04 16:14:23.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5979762cf1994e0a3c3b1c79f76ee8ba89a5547a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5979762cf1994e0a3c3b1c79f76ee8ba89a5547a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6fd6dd7c13e71c85c6f0f2d0591b77b81151e8fe","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b11f99ad25a0e18431604c0076b0e084061d8917":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2016-03-07 09:48:14.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/5"}}},"commit":{"parents":[{"commit":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","subject":"Merge \"Dump metric exception text to logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38c3f883dbbfe9807dcc8f9233b8147cdb7a182c"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-03-07 09:47:30.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b11f99ad25a0e18431604c0076b0e084061d8917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b11f99ad25a0e18431604c0076b0e084061d8917"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"88790d5bf235e06d36d94b39c4a772c60dfcc3b3":{"kind":"REWORK","_number":6,"created":"2016-03-20 13:34:37.000000000","uploader":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"ref":"refs/changes/24/275624/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/6"}}},"commit":{"parents":[{"commit":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","subject":"Merge \"Dump metric exception text to logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38c3f883dbbfe9807dcc8f9233b8147cdb7a182c"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Lenny Verkhovsky","email":"lennyb@mellanox.com","date":"2016-03-20 13:34:27.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88790d5bf235e06d36d94b39c4a772c60dfcc3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88790d5bf235e06d36d94b39c4a772c60dfcc3b3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"31c58285227e1d205b577a980f20014190c3ca8c":{"kind":"REWORK","_number":7,"created":"2016-03-20 14:09:18.000000000","uploader":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"ref":"refs/changes/24/275624/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/7"}}},"commit":{"parents":[{"commit":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","subject":"Merge \"Dump metric exception text to logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38c3f883dbbfe9807dcc8f9233b8147cdb7a182c"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Lenny Verkhovsky","email":"lennyb@mellanox.com","date":"2016-03-20 14:08:40.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/31c58285227e1d205b577a980f20014190c3ca8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/31c58285227e1d205b577a980f20014190c3ca8c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3fdc0845917b02504a03081f070116c00a1925b4":{"kind":"REWORK","_number":8,"created":"2016-03-21 09:41:10.000000000","uploader":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"ref":"refs/changes/24/275624/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/8"}}},"commit":{"parents":[{"commit":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","subject":"Merge \"Dump metric exception text to logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38c3f883dbbfe9807dcc8f9233b8147cdb7a182c"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Lenny Verkhovsky","email":"lennyb@mellanox.com","date":"2016-03-21 09:40:22.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fdc0845917b02504a03081f070116c00a1925b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fdc0845917b02504a03081f070116c00a1925b4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5845f2edac564f4773dca84b20ccc0b297bf70b8":{"kind":"REWORK","_number":9,"created":"2016-03-21 13:21:09.000000000","uploader":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"ref":"refs/changes/24/275624/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/9"}}},"commit":{"parents":[{"commit":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","subject":"Merge \"Dump metric exception text to logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38c3f883dbbfe9807dcc8f9233b8147cdb7a182c"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Lenny Verkhovsky","email":"lennyb@mellanox.com","date":"2016-03-21 13:19:16.000000000","tz":120},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5845f2edac564f4773dca84b20ccc0b297bf70b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5845f2edac564f4773dca84b20ccc0b297bf70b8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38c3f883dbbfe9807dcc8f9233b8147cdb7a182c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9abeeba46e02795d8a8a3d9ee25c9831eb3a4340":{"kind":"REWORK","_number":10,"created":"2016-03-30 07:30:30.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/10"}}},"commit":{"parents":[{"commit":"281ed64d1a29c27e13500d68a49f44e6049fbdbf","subject":"Merge \"Cleanup the exception LiveMigrationWithOldNovaNotSafe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/281ed64d1a29c27e13500d68a49f44e6049fbdbf"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-03-30 07:16:08.000000000","tz":180},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9abeeba46e02795d8a8a3d9ee25c9831eb3a4340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9abeeba46e02795d8a8a3d9ee25c9831eb3a4340"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"281ed64d1a29c27e13500d68a49f44e6049fbdbf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"615a2ce3448264db3a6eb6d58a6d4c05212d7a57":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2016-04-04 15:53:26.000000000","uploader":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"ref":"refs/changes/24/275624/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/11"}}},"commit":{"parents":[{"commit":"b35a1105bf7b275fd43377dd17db0ce56fbf0b7f","subject":"Merge \"Add Service.get_minimum_version_multi() for multiple binaries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b35a1105bf7b275fd43377dd17db0ce56fbf0b7f"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Lenny Verkhovsky","email":"lennyb@mellanox.com","date":"2016-04-04 15:53:25.000000000","tz":0},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/615a2ce3448264db3a6eb6d58a6d4c05212d7a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/615a2ce3448264db3a6eb6d58a6d4c05212d7a57"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b35a1105bf7b275fd43377dd17db0ce56fbf0b7f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d9ee4a6001dc87a8474f3329918decffd2d44bab":{"kind":"REWORK","_number":12,"created":"2016-05-23 16:38:21.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/12"}}},"commit":{"parents":[{"commit":"2505c5d8b1d9c075e20275ee903657640cc97c92","subject":"Merge \"Hyper-V: Implement nova rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2505c5d8b1d9c075e20275ee903657640cc97c92"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-05-23 16:32:54.000000000","tz":180},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9ee4a6001dc87a8474f3329918decffd2d44bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9ee4a6001dc87a8474f3329918decffd2d44bab"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2505c5d8b1d9c075e20275ee903657640cc97c92","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fd9f8e4255a0994af35c9553969c627034fb37af":{"kind":"REWORK","_number":13,"created":"2016-05-24 09:34:04.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/13"}}},"commit":{"parents":[{"commit":"2505c5d8b1d9c075e20275ee903657640cc97c92","subject":"Merge \"Hyper-V: Implement nova rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2505c5d8b1d9c075e20275ee903657640cc97c92"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-05-24 09:30:52.000000000","tz":180},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd9f8e4255a0994af35c9553969c627034fb37af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd9f8e4255a0994af35c9553969c627034fb37af"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2505c5d8b1d9c075e20275ee903657640cc97c92","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3b28ae8e20c6f1c883dea518e97d2c937139cc9f":{"kind":"REWORK","_number":14,"created":"2016-05-30 09:39:52.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/14"}}},"commit":{"parents":[{"commit":"2505c5d8b1d9c075e20275ee903657640cc97c92","subject":"Merge \"Hyper-V: Implement nova rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2505c5d8b1d9c075e20275ee903657640cc97c92"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-05-30 09:37:25.000000000","tz":180},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b28ae8e20c6f1c883dea518e97d2c937139cc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b28ae8e20c6f1c883dea518e97d2c937139cc9f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2505c5d8b1d9c075e20275ee903657640cc97c92","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a27481dd7659b9c2a29f73a9d4da90ae4db65457":{"kind":"REWORK","_number":15,"created":"2016-09-19 16:55:39.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/15"}}},"commit":{"parents":[{"commit":"2505c5d8b1d9c075e20275ee903657640cc97c92","subject":"Merge \"Hyper-V: Implement nova rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2505c5d8b1d9c075e20275ee903657640cc97c92"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-19 09:53:56.000000000","tz":180},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a27481dd7659b9c2a29f73a9d4da90ae4db65457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a27481dd7659b9c2a29f73a9d4da90ae4db65457"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2505c5d8b1d9c075e20275ee903657640cc97c92","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5897f039c857499d03a4722faebef7aed6521ef9":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2016-11-14 11:09:21.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/24/275624/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/275624/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/275624/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/275624/16"}}},"commit":{"parents":[{"commit":"3b0c53e136f7b430cc13f35d2b91125008816d0f","subject":"Merge \"Revert MTU hacks for bug 1623876\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b0c53e136f7b430cc13f35d2b91125008816d0f"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2015-10-12 15:19:02.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-11-14 11:09:21.000000000","tz":0},"subject":"[WIP] OVS offload POC","message":"[WIP] OVS offload POC\n\nChange-Id: I39790799e3d278c96e022b6ec04c310bab3b6895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5897f039c857499d03a4722faebef7aed6521ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5897f039c857499d03a4722faebef7aed6521ef9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3b0c53e136f7b430cc13f35d2b91125008816d0f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
