)]}'
{"id":"openstack%2Fnova~275800","triplet_id":"openstack%2Fnova~master~I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c","project":"openstack/nova","branch":"master","topic":"bug/1489853","hashtags":[],"change_id":"I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c","subject":"libvirt: prevent serial console port exhaustion","status":"ABANDONED","created":"2016-02-03 16:25:52.000000000","updated":"2016-04-06 13:53:15.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"031141157fed784b8d1010d599bc04007efa67f1","_number":275800,"virtual_id_number":275800,"owner":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"all":[{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},{"date":"2016-03-15 15:54:15.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},{"value":1,"date":"2016-03-15 14:32:54.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-03-15 17:05:04.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-03-15 13:29:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},{"date":"2016-03-15 12:50:36.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-15 12:43:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},{"date":"2016-03-15 16:03:19.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-15 13:46:10.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"date":"2016-03-15 14:04:41.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-15 13:45:15.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"disliked":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"all":[{"value":1,"date":"2016-03-15 12:23:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2016-03-15 12:23:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-03-31 14:48:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},{"value":1,"date":"2016-03-15 12:23:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"all":[{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2016-03-31 14:49:55.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-03 16:25:53.000000000","updated_by":{"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},"reviewer":{"_account_id":14594,"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","username":"alexs"},"state":"REVIEWER"},{"updated":"2016-02-08 19:55:58.000000000","updated_by":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"reviewer":{"_account_id":12299,"name":"Paweł Koniszewski","email":"pawel.koniszewski@gmail.com","username":"pawel.koniszewski"},"state":"REVIEWER"},{"updated":"2016-02-18 02:18:01.000000000","updated_by":{"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},"reviewer":{"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},"state":"REVIEWER"},{"updated":"2016-02-23 12:52:47.000000000","updated_by":{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},"reviewer":{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},"state":"REVIEWER"},{"updated":"2016-02-23 14:27:04.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-08 10:23:50.000000000","updated_by":{"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},"reviewer":{"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},"state":"REVIEWER"},{"updated":"2016-03-08 16:31:45.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2016-03-09 11:56:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2016-03-14 09:05:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-14 22:13:30.000000000","updated_by":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"reviewer":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"state":"REVIEWER"},{"updated":"2016-03-15 12:43:34.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 12:50:36.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 13:29:07.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 13:45:15.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 13:46:10.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 14:04:41.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 14:32:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-03-15 15:54:15.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 16:03:19.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-15 17:05:04.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-03-31 14:48:39.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"}],"messages":[{"id":"329b6cf68a715b3833931227d177c05910b75593","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-03 16:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aed1ba3cb595c6c99baec24d4fbaa9bcbec93421","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-03 16:26:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"38c4497ce415de72df7eb12ec0f1ca2fe949fdfc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-03 17:23:55.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/1/check/check-ibm-tempest-dsvm-full/9054ac4/ : SUCCESS in 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"204649600db75a03b8b3b0945e7f4d0175d7ee19","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-03 17:35:55.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/275800/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b6e383ffc30440cd312bdced2dd2c7f5cbc6fa24","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-03 17:41:29.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6f2a32c76d8d4bbba87f7b51d463ff4b/index.html : SUCCESS in 11m 48s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/dd955de218ef47fe87eff361550f891a/index.html : SUCCESS in 12m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/2829ee0ff6ef43d697c4f4d44309490d/index.html : SUCCESS in 14m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/1/check/real-db-upgrade_nova_percona_user_001:th-percona/b9c86ca4df4a47fcb53b83bb331f921f/index.html : SUCCESS in 13m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/200c3ea72c58433b9daae99083170e21/index.html : SUCCESS in 14m 05s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/1/check/real-db-upgrade_nova_percona_user_002:th-percona/64aa5172fd5c461b886e0acaa4da5190/index.html : SUCCESS in 14m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"199e5a39c45dbe4a7e868a1409a378753879d5c0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-03 18:44:56.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/1/5140/index.html : SUCCESS in 01 h 22 m 17 s","accounts_in_message":[],"_revision_number":1},{"id":"cdf8878c7deac289f85b927d1ddd2c1d2b81399b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-03 19:48:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/00/275800/1/check/dsvm-tempest-xen/06b059a : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c56cda1740610b6ef46ba32c96d391ce981fda54","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-03 20:32:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/1/check/gate-nova-pep8/018893b/ : SUCCESS in 7m 38s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/1/check/gate-nova-docs/02728be//doc/build/html/ : SUCCESS in 11m 40s\n- gate-nova-python27 http://logs.openstack.org/00/275800/1/check/gate-nova-python27/43021ba/ : SUCCESS in 14m 44s\n- gate-nova-python34 http://logs.openstack.org/00/275800/1/check/gate-nova-python34/d7a88f6/ : SUCCESS in 27m 31s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-full/36b6444/ : SUCCESS in 46m 15s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-postgres-full/f1a05aa/ : SUCCESS in 1h 01m 48s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-neutron-full/c0707f9/ : FAILURE in 1h 22m 43s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/1/check/gate-grenade-dsvm/d9f0d47/ : SUCCESS in 56m 04s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-large-ops/8ecdd2b/ : SUCCESS in 31m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-neutron-large-ops/f6dcdc7/ : SUCCESS in 40m 20s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/1/check/gate-nova-tox-functional/7b0941b/ : SUCCESS in 16m 44s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/1/check/gate-grenade-dsvm-multinode/efdfc49/ : FAILURE in 50m 34s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2bd5423/ : SUCCESS in 41m 13s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-nova-v20-api/4a19c2f/ : SUCCESS in 39m 32s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-nova-v20-api-legacy/c394a18/ : SUCCESS in 43m 04s\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-cells/5d54b01/ : SUCCESS in 36m 56s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-full-ceph/3b25000/ : SUCCESS in 44m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/bc8ec38/ : SUCCESS in 43m 19s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/1/check/gate-tempest-dsvm-multinode-full/745bec7/ : FAILURE in 1h 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45503ab720603f91a3162896e89b54757f1080dd","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-03 23:36:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/1/ : SUCCESS in 2h 54m 10s","accounts_in_message":[],"_revision_number":1},{"id":"837d1137b54e779eef7074de7a5aa00f0031dbba","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 02:48:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/1/check/check-dsvm-tempest-vz7-exe-minimal/c7d0115 : SUCCESS in 2h 06m 58s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"2f59c5726d4a6c77b5ea2394ca917b3130e3070f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 04:27:59.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/00/275800/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"4fc190de72d7d20008e58b9cc3d72721d1f826fa","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-04 16:12:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94bc1ef2831d99af9e9aa93c23083624f6187a23","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-04 16:28:23.000000000","message":"Patch Set 2: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/2/5293/index.html : FAILURE in 14 m 01 s","accounts_in_message":[],"_revision_number":2},{"id":"8b3889e15f61348e69d5a7ae2c2f7a202e098ef6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-04 16:34:13.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/275800/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"7daa547be1480efdc5948ef1f5f904fce9d98b49","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-04 16:36:07.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f26d765b632e4d5183a9ee0017048a88/index.html : SUCCESS in 14m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/295e43bbf73d4d9f81248063012efc46/index.html : SUCCESS in 15m 45s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0a9bd1137ffa45ddaf57fa1e2495860b/index.html : SUCCESS in 14m 57s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/2/check/real-db-upgrade_nova_percona_user_001:th-percona/94071e9b2232450b9a83490e0d5e64cf/index.html : SUCCESS in 16m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/05e96594748c4b6e886f98964142c12d/index.html : SUCCESS in 14m 38s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/2/check/real-db-upgrade_nova_percona_user_002:th-percona/7bfa0079a4964cac947ce250a3884d9f/index.html : SUCCESS in 14m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"9d7a0e7aefd464574224fdc3d8eca3d810ca4bb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-04 17:13:04.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/2/check/check-ibm-tempest-dsvm-full/caefca1/ : SUCCESS in 1h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e62b3b39b761f6a8152e97c9301574f1b0dc7e0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 18:54:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/2/ : SUCCESS in 2h 19m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7953e618d0945c89b6f60ccf6366cc5fd5123877","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-04 19:26:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/2/check/check-dsvm-tempest-vz7-exe-minimal/34665c1 : SUCCESS in 1h 02m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"38c392e8eb71a078b984c4f6ebb7fd9a8662a906","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-04 19:33:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/2/check/gate-nova-pep8/6de6385/ : SUCCESS in 9m 13s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/2/check/gate-nova-docs/404c357//doc/build/html/ : SUCCESS in 14m 17s\n- gate-nova-python27 http://logs.openstack.org/00/275800/2/check/gate-nova-python27/fa849f4/ : SUCCESS in 15m 05s\n- gate-nova-python34 http://logs.openstack.org/00/275800/2/check/gate-nova-python34/9fe5bc6/ : SUCCESS in 27m 58s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-full/5f090f5/ : SUCCESS in 47m 42s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-postgres-full/ee63b81/ : SUCCESS in 1h 03m 15s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-neutron-full/7c4f07c/ : SUCCESS in 1h 01m 20s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/2/check/gate-grenade-dsvm/6685224/ : SUCCESS in 46m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-large-ops/5c0ffee/ : SUCCESS in 43m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-neutron-large-ops/c4af57f/ : SUCCESS in 42m 13s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/2/check/gate-nova-tox-functional/4008d12/ : SUCCESS in 17m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/2/check/gate-grenade-dsvm-multinode/b299f15/ : SUCCESS in 1h 00m 45s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f7e4320/ : SUCCESS in 41m 05s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-nova-v20-api/7458938/ : SUCCESS in 39m 00s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-nova-v20-api-legacy/6b5654c/ : SUCCESS in 37m 22s\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-cells/2130f9b/ : SUCCESS in 39m 20s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-full-ceph/6f88516/ : SUCCESS in 58m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/6fd2298/ : SUCCESS in 44m 59s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/2/check/gate-tempest-dsvm-multinode-full/364172f/ : FAILURE in 1h 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c9c5758ed1343293dadf3b0e8c6f7607bb30f6a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-04 20:09:24.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-275800-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"a3f2452f5b2726ae98f34b5a1cbc435a543c1f7b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-04 23:55:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275800/2/Nova-ML2-Sriov : SUCCESS in 42m 38s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c3a891e5c5b6290e87f72b8c6b98d508d9aba18","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-05 08:39:03.000000000","message":"Patch Set 2:\n\nxenserver: recheck","accounts_in_message":[],"_revision_number":2},{"id":"ea9c81f5a402c51ae253794becfe2ae9a2dc0044","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-05 09:55:14.000000000","message":"Patch Set 2:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/00/275800/2 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"331146d75896d2c7b1d20ea06f31e7263b8936c0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-05 10:24:20.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/2/5357/index.html : SUCCESS in 01 h 31 m 02 s","accounts_in_message":[],"_revision_number":2},{"id":"de9018832a29a937f91e674bf14a579b35073e9f","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-08 10:40:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d93df612e2fa4eb6b8e47adfdc603048ba3fc71a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-08 10:58:28.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/275800/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"1818b56296ff3d964899f8bfc782c316201d25ac","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-08 10:59:28.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/463e5866fda7473f927f89751d64cd78/index.html : SUCCESS in 12m 30s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/635fc26dc8fa4a49ac86ff72866be489/index.html : SUCCESS in 15m 12s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/3032ec77f285470cbe13c9021a2b0be0/index.html : SUCCESS in 11m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_percona_user_001:th-percona/e747b925f30b47b99817e206aa1c33a1/index.html : SUCCESS in 15m 49s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a641a94bc75849709ca342531243db32/index.html : SUCCESS in 17m 10s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_percona_user_002:th-percona/a4c3f8fbcbb54ac8b852cfa3c00c6738/index.html : SUCCESS in 12m 12s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"39e4148b6161d4f93cac8b3ed90832ba196d0304","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-08 11:41:33.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/3/check/check-ibm-tempest-dsvm-full/b61d743/ : FAILURE in 1h 00m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f111a0590508535b4d0eac7eb5f25c5798c341c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-08 11:54:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275800/3/Nova-ML2-Sriov : SUCCESS in 41m 39s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"b63c19caea2280b7fa4e1c7cee3d01372cd22a91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-08 12:01:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/3/check/gate-nova-pep8/45dbe3b/ : SUCCESS in 3m 35s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/3/check/gate-nova-docs/c224768//doc/build/html/ : SUCCESS in 9m 57s\n- gate-nova-python27 http://logs.openstack.org/00/275800/3/check/gate-nova-python27/6886716/ : SUCCESS in 10m 46s\n- gate-nova-python34 http://logs.openstack.org/00/275800/3/check/gate-nova-python34/4d47ab0/ : SUCCESS in 28m 22s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full/c9a278e/ : SUCCESS in 35m 56s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-postgres-full/bbd241e/ : SUCCESS in 46m 32s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-neutron-full/4573250/ : SUCCESS in 1h 17m 38s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/3/check/gate-grenade-dsvm/25794b3/ : SUCCESS in 46m 24s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-large-ops/6b2f0f2/ : SUCCESS in 37m 47s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-neutron-large-ops/4b49691/ : SUCCESS in 47m 07s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/3/check/gate-nova-tox-functional/ade85b2/ : SUCCESS in 11m 36s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/3/check/gate-grenade-dsvm-multinode/9c71906/ : SUCCESS in 41m 21s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/89970cd/ : FAILURE in 59m 43s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-nova-v20-api/0145a08/ : SUCCESS in 57m 34s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-nova-v20-api-legacy/c191a42/ : SUCCESS in 27m 53s\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-cells/c2ec12e/ : FAILURE in 30m 14s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full-ceph/0af5ae5/ : FAILURE in 41m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/89313bc/ : FAILURE in 42m 29s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-multinode-full/4473308/ : SUCCESS in 58m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f4238a2650253096dcf8d475dcb12c1eb4be3a6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-08 12:20:39.000000000","message":"Patch Set 3: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/3/5510/index.html : FAILURE in 01 h 33 m 54 s","accounts_in_message":[],"_revision_number":3},{"id":"76a9ef359950f8a8b07f4661f111f4aa3f461bd5","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-08 12:45:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"992a755aa8afc1a758cfe3f5c022ceaf35e373ec","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-08 13:25:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/3/check/check-dsvm-tempest-vz7-exe-minimal/9cdf950 : SUCCESS in 1h 15m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"43e4b94f633fb5a351ac2755679e2d1bc1d463e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-08 13:46:09.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/3/check/check-ibm-tempest-dsvm-full/5d70ef3/ : FAILURE in 1h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2d783299f046f76c70bc92b42b7bd043034bf43","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-08 13:46:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/3/check/gate-nova-pep8/450f981/ : SUCCESS in 4m 19s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/3/check/gate-nova-docs/084177e//doc/build/html/ : SUCCESS in 9m 50s\n- gate-nova-python27 http://logs.openstack.org/00/275800/3/check/gate-nova-python27/ea14dd5/ : SUCCESS in 11m 22s\n- gate-nova-python34 http://logs.openstack.org/00/275800/3/check/gate-nova-python34/8bfb6f9/ : SUCCESS in 29m 31s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full/6692957/ : SUCCESS in 42m 13s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-postgres-full/13ccbb7/ : SUCCESS in 40m 09s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-neutron-full/be46406/ : SUCCESS in 48m 35s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/3/check/gate-grenade-dsvm/f9d06f6/ : SUCCESS in 40m 36s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-large-ops/344b232/ : SUCCESS in 28m 10s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-neutron-large-ops/fe13cee/ : SUCCESS in 27m 57s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/3/check/gate-nova-tox-functional/2bedd60/ : SUCCESS in 12m 22s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/3/check/gate-grenade-dsvm-multinode/4db91a1/ : SUCCESS in 44m 09s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/832cc4b/ : SUCCESS in 36m 24s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-nova-v20-api/105f328/ : SUCCESS in 33m 19s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-nova-v20-api-legacy/129faab/ : SUCCESS in 31m 30s\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-cells/9e69664/ : FAILURE in 33m 22s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full-ceph/8d9956a/ : SUCCESS in 40m 01s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/cca43f8/ : SUCCESS in 51m 28s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-multinode-full/95516d8/ : SUCCESS in 54m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a86956c91e05a6c122e06dbdb1e1f8d1896457a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-08 14:09:45.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/3/ : FAILURE in 3h 28m 13s","accounts_in_message":[],"_revision_number":3},{"id":"d995053622dab77714b5c6400037fa1cc27153a1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-08 15:41:11.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/00/275800/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"6d912fd7c661b74aefa9bcc41b1759c1418cddcc","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-08 17:15:22.000000000","message":"Patch Set 3:\n\nrecheck bug 1543106","accounts_in_message":[],"_revision_number":3},{"id":"d973baf04d7d51cf4252aee9f8bccd47555bdaab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-08 19:07:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/3/check/gate-nova-pep8/ff8dc01/ : SUCCESS in 4m 35s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/3/check/gate-nova-docs/1ff8e29//doc/build/html/ : SUCCESS in 7m 31s\n- gate-nova-python27 http://logs.openstack.org/00/275800/3/check/gate-nova-python27/f18255f/ : SUCCESS in 10m 46s\n- gate-nova-python34 http://logs.openstack.org/00/275800/3/check/gate-nova-python34/90af51b/ : SUCCESS in 31m 14s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full/fc17e81/ : SUCCESS in 56m 01s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-postgres-full/b31ec56/ : SUCCESS in 42m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-neutron-full/623b465/ : SUCCESS in 51m 40s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/3/check/gate-grenade-dsvm/b84e26d/ : SUCCESS in 42m 55s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-large-ops/612fccf/ : SUCCESS in 32m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-neutron-large-ops/93a4b76/ : SUCCESS in 32m 36s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/3/check/gate-nova-tox-functional/849d4e5/ : SUCCESS in 12m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/3/check/gate-grenade-dsvm-multinode/44b3874/ : SUCCESS in 52m 06s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/aa9720e/ : SUCCESS in 34m 24s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-nova-v20-api/bf7f5b4/ : SUCCESS in 46m 57s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-nova-v20-api-legacy/d425402/ : SUCCESS in 46m 07s\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-cells/8a3bb1f/ : FAILURE in 34m 49s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full-ceph/da5956e/ : FAILURE in 44m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/0d92cd5/ : SUCCESS in 53m 15s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/3/check/gate-tempest-dsvm-multinode-full/a74b89c/ : FAILURE in 1h 00m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d814d9b32ce4440d296bc72fbf08d5c51fa63bc9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-08 19:23:17.000000000","message":"Patch Set 3:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0374e0a0413349468cf7b70a259e8376/index.html : SUCCESS in 14m 13s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4bf55458c2d340228b5a1b7a574bdb96/index.html : SUCCESS in 16m 14s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/521a4275accf4e42ae4b24be03f26750/index.html : SUCCESS in 15m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_percona_user_001:th-percona/00a8e06919f6499ba6691112edb90998/index.html : SUCCESS in 15m 54s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d8ec3cf3afab4a82ba9044299efbb9c6/index.html : SUCCESS in 14m 55s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/3/check/real-db-upgrade_nova_percona_user_002:th-percona/b19e030bc9124713ac80b0e7b904a290/index.html : SUCCESS in 14m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"a455478ffee34aa0c6cc4af0aaa32faadbb82667","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-08 22:00:27.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/3/ : FAILURE in 3h 07m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0960d9b7374698c14b705707a4d348cce125effa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-09 01:57:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275800/3/Nova-ML2-Sriov : SUCCESS in 34m 21s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"772919f70ca607a06892d3cbe36afaf78f603577","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-02-09 10:51:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d8790a5ed332e2e2b6d572c49469eb81d34446c1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-09 11:09:00.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/275800/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"5664d60edc4aa105d3dc6d33ae45fdab33caf4cd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-09 11:10:00.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6b5e6238f9ee4fe2b76fbc14151addee/index.html : SUCCESS in 15m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ed1a95fb9ef949c38b1e66f37731372d/index.html : SUCCESS in 17m 09s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/cc4420a240ae47d08d60e41e98d00838/index.html : SUCCESS in 14m 39s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/4/check/real-db-upgrade_nova_percona_user_001:th-percona/f91d7ef07e4d40a483a5f37efb19729e/index.html : SUCCESS in 13m 24s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/56af2c661e214c3996c3289a22588c33/index.html : SUCCESS in 15m 15s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/4/check/real-db-upgrade_nova_percona_user_002:th-percona/dff545b34c644588a6dbc9d59677f5f7/index.html : SUCCESS in 15m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"5c6c8b1aa997205faa9bfc2034eac04e691be385","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-09 11:53:07.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/4/check/check-ibm-tempest-dsvm-full/dd36e7b/ : SUCCESS in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"713baf37d9fffcd07e025884720a8a61d3dc569a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-09 12:08:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/00/275800/4/check/dsvm-tempest-xen/5820db0 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":4},{"id":"535969619dd1d82643c0271c27933a1918a07347","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-09 12:09:38.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/4/5657/index.html : SUCCESS in 01 h 14 m 23 s","accounts_in_message":[],"_revision_number":4},{"id":"f70a5089925267d04297ad200a018817de75dea0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-09 12:26:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/4/check/gate-nova-pep8/970f5d1/ : SUCCESS in 2m 50s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/4/check/gate-nova-docs/d575d8d//doc/build/html/ : SUCCESS in 7m 23s\n- gate-nova-python27 http://logs.openstack.org/00/275800/4/check/gate-nova-python27/5f76c34/ : SUCCESS in 9m 04s\n- gate-nova-python34 http://logs.openstack.org/00/275800/4/check/gate-nova-python34/31b96af/ : SUCCESS in 27m 18s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-full/9b55cb9/ : SUCCESS in 39m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-postgres-full/990563f/ : SUCCESS in 32m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-neutron-full/ede0883/ : SUCCESS in 1h 29m 36s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/4/check/gate-grenade-dsvm/993e825/ : SUCCESS in 34m 07s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-large-ops/622ade0/ : SUCCESS in 35m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-neutron-large-ops/03b44a3/ : SUCCESS in 21m 22s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/4/check/gate-nova-tox-functional/08dd3ae/ : SUCCESS in 12m 03s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/4/check/gate-grenade-dsvm-multinode/1a18ea4/ : SUCCESS in 39m 56s (non-voting)\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/3a563ab/ : FAILURE in 43m 17s (non-voting)\n- gate-tempest-dsvm-nova-v20-api http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-nova-v20-api/5f7d9da/ : SUCCESS in 44m 05s\n- gate-tempest-dsvm-nova-v20-api-legacy http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-nova-v20-api-legacy/d7b88df/ : SUCCESS in 33m 34s\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-cells/a35016c/ : SUCCESS in 55m 51s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-full-ceph/66c2854/ : SUCCESS in 36m 06s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/1bbaa9b/ : SUCCESS in 34m 29s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/4/check/gate-tempest-dsvm-multinode-full/aa12069/ : FAILURE in 58m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9992090ce650c4aa5cf0ca6c4e6acdf944c74016","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-09 12:26:30.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-275800-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"af0e89eb6e1c4966a4e98c14ddfa9647de68d149","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 12:35:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/4/check/check-dsvm-tempest-vz7-exe-minimal/498a824 : SUCCESS in 1h 42m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"a766c43d3211eb5563be3b0d8f4bb4fffd4eb207","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-09 13:26:30.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/4/ : FAILURE in 2h 33m 52s","accounts_in_message":[],"_revision_number":4},{"id":"fd88b16bb070cebe9ef0084a4e1bc28c2a1b5a11","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-14 00:09:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275800/4/2198 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"8e957f52ebed9e9cedb58325e8a423feb94dbed5","author":{"_account_id":16366,"name":"Randall Nortman","email":"openstack@nortman.net","username":"rnortman"},"date":"2016-02-18 02:18:01.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nCode appears correct but I have some nits.","accounts_in_message":[],"_revision_number":4},{"id":"dc27b5db958bb7917b9d160bb48e4f27c488ea78","author":{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},"date":"2016-02-23 12:52:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"343666d325e804c7cfc7d167d6d95d7df359c23f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-23 13:10:19.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/275800/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"ff43837b1a03511b7101847867068c172713a9f9","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-23 13:12:02.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f2bc7776f9594fc1bf1b4ac55a19fbf6/index.html : SUCCESS in 12m 49s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a626406b41084b459f71ca0444aa7ce5/index.html : SUCCESS in 15m 47s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/75d14ad88b0b4496b0a136fb11e39aa0/index.html : SUCCESS in 17m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/5/check/real-db-upgrade_nova_percona_user_001:th-percona/1933393d2405430b8a564ad61668ee74/index.html : SUCCESS in 13m 23s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/65d9e6d6e5824ea18ad7c52033945b27/index.html : SUCCESS in 14m 34s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/5/check/real-db-upgrade_nova_percona_user_002:th-percona/768a7dd3fbd54c28ac98aa2ee3eeb249/index.html : SUCCESS in 14m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"68dab860c1c1d358837727f184254f8a0b096046","author":{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},"date":"2016-02-23 13:23:00.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2908c4cfe73864d3f3e9b760aa720ceab3535406","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-23 13:49:54.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/5/check/check-ibm-tempest-dsvm-full/af0c76d/ : SUCCESS in 55m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"485e81613e7f534c1adc0f17996ecf1d952e2edb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-23 13:56:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/5/check/gate-nova-pep8/b2b3622/ : SUCCESS in 2m 16s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/5/check/gate-nova-docs/fbf8848//doc/build/html/ : SUCCESS in 6m 33s\n- gate-nova-python27 http://logs.openstack.org/00/275800/5/check/gate-nova-python27/bf09bda/ : SUCCESS in 9m 37s\n- gate-nova-python34 http://logs.openstack.org/00/275800/5/check/gate-nova-python34/660fa53/ : SUCCESS in 28m 27s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-full/7b6686b/ : SUCCESS in 35m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-postgres-full/1c257cb/ : SUCCESS in 47m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-neutron-full/a7ca691/ : SUCCESS in 1h 02m 12s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/5/check/gate-grenade-dsvm/40a7f62/ : SUCCESS in 36m 51s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/5/check/gate-nova-tox-functional/1a1fc45/ : SUCCESS in 9m 37s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/5/check/gate-grenade-dsvm-multinode/618773c/ : SUCCESS in 38m 23s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/881c3a6/ : SUCCESS in 47m 07s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-cells/70fb66f/ : SUCCESS in 30m 53s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-full-ceph/28bd3bf/ : SUCCESS in 39m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/d020a06/ : SUCCESS in 34m 15s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/5/check/gate-tempest-dsvm-multinode-full/090875b/ : FAILURE in 46m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"304ffb5c20ac7a57723730d9ffdbef3f34d10957","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-02-23 13:57:16.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-275800-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"ee3e80c2237c523f7b07b6810a737bfb2bb10117","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-23 14:06:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/00/275800/5/check/dsvm-tempest-xen/8b620ea : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":5},{"id":"20c6426f5814e6649701f8ff49c17c5ef5d13414","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-23 14:07:50.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/5/7020/index.html : SUCCESS in 01 h 12 m 41 s","accounts_in_message":[],"_revision_number":5},{"id":"b81cd86bf93086b695bf2343b2a872713c3a85e5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-23 14:27:04.000000000","message":"Patch Set 5:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/00/275800/5 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"d9e59cdcd193ebeb5785ee43f9bf252c25348388","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-23 14:56:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/5/ : SUCCESS in 2h 02m 26s","accounts_in_message":[],"_revision_number":5},{"id":"084923362bb1f90cdeb3955ed35e82fbb374f664","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-23 15:10:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/5/check/check-dsvm-tempest-vz7-exe-minimal/ccb9e2b : SUCCESS in 56m 01s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"9df359176fc529a7b9289d78c0c0edbf4b12fea3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-02-23 16:21:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275800/5/Nova-ML2-Sriov : SUCCESS in 43m 59s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"273df00e82b71e63bce6664dd10c79134ef05174","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-25 09:44:11.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275800/5/2557 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"a67efb670bbcee48014b1d1da9b62ea223e53b1e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 07:31:44.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"c29643b6063a3b930f0b0d9bf7f154f61e1a2530","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-08 09:35:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"223f844cac9d7351200ed3ab1ae52131c291cad8","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-08 09:55:21.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/22c4389e609341329ecf79a9a6ca4c29/index.html : SUCCESS in 13m 38s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f5be3057b2be45f7b5f44044864bb6d5/index.html : SUCCESS in 18m 32s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/db164f88b4e84858b136cf59f1e4c487/index.html : SUCCESS in 15m 10s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/6/check/real-db-upgrade_nova_percona_user_001:th-percona/10934fbd40f64d7a8b0fc0d186862f1e/index.html : SUCCESS in 16m 48s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/3886970565a349a08277fe8d569a3397/index.html : SUCCESS in 12m 46s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/6/check/real-db-upgrade_nova_percona_user_002:th-percona/6cc383b28d9646de91821f3419d29d48/index.html : SUCCESS in 15m 20s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"761f40bb657c0492b69b1b6d4a554adbffaa8f4c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-08 09:59:22.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/275800/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"1afd94150ddd1703fd75af6f4b22969e678100ef","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-08 10:40:26.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/6/check/check-ibm-tempest-dsvm-full/b45abc2/ : SUCCESS in 1h 04m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3a225180abafd4a3da9a61f8622d026326584523","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-03-08 10:53:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275800/6/Nova-ML2-Sriov : SUCCESS in 50m 13s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"6a42148e8d1abe1e4361c2605ea21a41732155ad","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-08 10:55:05.000000000","message":"Patch Set 6: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/6/8619/index.html : SUCCESS in 01 h 13 m 30 s","accounts_in_message":[],"_revision_number":6},{"id":"fc4190d177342fed744f457dd4f75186c8b3a99a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-08 10:59:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/6/check/gate-nova-pep8/3ed26d4/ : SUCCESS in 5m 40s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/6/check/gate-nova-docs/1e75fc3//doc/build/html/ : SUCCESS in 8m 30s\n- gate-nova-python27 http://logs.openstack.org/00/275800/6/check/gate-nova-python27/1e78716/ : SUCCESS in 11m 58s\n- gate-nova-python34 http://logs.openstack.org/00/275800/6/check/gate-nova-python34/fbf1414/ : SUCCESS in 30m 37s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-full/e10a86a/ : SUCCESS in 59m 54s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-postgres-full/478a0ee/ : SUCCESS in 40m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-neutron-full/81f97d2/ : SUCCESS in 55m 20s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/6/check/gate-grenade-dsvm/6646078/ : SUCCESS in 39m 57s\n- gate-nova-tox-functional http://logs.openstack.org/00/275800/6/check/gate-nova-tox-functional/d650e18/ : SUCCESS in 13m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/6/check/gate-grenade-dsvm-multinode/1f5d2a8/ : SUCCESS in 43m 09s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c265192/ : SUCCESS in 33m 14s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-cells/58a013e/ : SUCCESS in 36m 24s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-full-ceph/de6eab2/ : SUCCESS in 1h 20m 50s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/4e5c05d/ : SUCCESS in 1h 21m 19s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/6/check/gate-tempest-dsvm-multinode-full/48fd79c/ : SUCCESS in 48m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9ca2d9f212400d5577164de778e44e5f454b618d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-08 11:04:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/00/275800/6/check/dsvm-tempest-xen/f2a0a9d : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":6},{"id":"38062aa4361981f402a1962e9289b9cbecf91827","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-08 12:50:33.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/6/ : FAILURE in 53m 35s","accounts_in_message":[],"_revision_number":6},{"id":"831adb62f8177598bb1c2f7428f5a2fc8c8cb62b","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2016-03-08 16:31:45.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"921699e390d87de31eccf3f649f50d7e1a69327e","author":{"_account_id":20417,"name":"GuoChang Tang","email":"tangguochang@139.com"},"date":"2016-03-09 00:55:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ffde4df94af6fa25c638497d6c726389a3e82b31","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-09 02:44:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/6/check/check-dsvm-tempest-vz7-exe-minimal/7405604 : SUCCESS in 1h 16m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"9f462fc909f4bb787a35dd1c955a0f935654ec17","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-09 10:50:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bbaaaf537b743733bc9910c25c52c77020d167a4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-03-09 11:56:20.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2ca18f21e785eacef4171ec90f332574a839ae63","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2016-03-09 12:34:23.000000000","message":"Patch Set 6: Code-Review-1\n\nCode looks good for me, just question. This can create race, if an other instance is created between that time of rebuild, no?  Also why do we care about to reuse same ports since when we destroy the instance, ports will be released. I think it\u0027s a lot of complicated code form something that need (or not) be fixed to the method responsible to get ports from \"bucket\".","accounts_in_message":[],"_revision_number":6},{"id":"bdfafed260bfb834fd69b6bdac26eb039feab948","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2016-03-09 12:48:16.000000000","message":"Patch Set 6:\n\nWhy I do not think it\u0027s good idea it\u0027s because we have thread-safe methods to release and acquire ports from a bucket but by your code we complete avoid them to make it less safe.","accounts_in_message":[],"_revision_number":6},{"id":"78ca4f1ab3fa295e89af15e82d4ed85e9caf3c7f","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-09 15:15:37.000000000","message":"Patch Set 6:\n\n\u003e Code looks good for me, just question. This can create race, if an\n \u003e other instance is created between that time of rebuild, no?  Also\n \u003e why do we care about to reuse same ports since when we destroy the\n \u003e instance, ports will be released. I think it\u0027s a lot of complicated\n \u003e code form something that need (or not) be fixed to the method\n \u003e responsible to get ports from \"bucket\".\n\n@sahid:\n\nThe base for this patch series is [1] from Alex. Unfortunately he left\nthe team. I split up [1] into 3 changes each with a specific \nresponsibility to make a review easier. The code for creating the\nserial console is called multiple times during a live migration.\nThe ports \"bucket\" doesn\u0027t know that context and allocates new ports \neach time. The ports \"bucket\" shouldn\u0027t know that context as it is not\nin its responsibility (or should it?).\n\nI\u0027m not sure if I understand the circumstances where a race condition \ncan occur. Could you explain that more pleas?\n\nReferences:\n[1] https://review.openstack.org/#/c/215102/12/nova/virt/libvirt/driver.py","accounts_in_message":[],"_revision_number":6},{"id":"2b42b781acd55d19bdd381ada65f7f76f87e27ff","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-09 15:18:24.000000000","message":"Patch Set 6:\n\n\u003e Why I do not think it\u0027s good idea it\u0027s because we have thread-safe\n \u003e methods to release and acquire ports from a bucket but by your code\n \u003e we complete avoid them to make it less safe.\n\n@sahid: \nDo you mean that the re-use is the way I avoid the thread-safe code from [1]?\n\nReferences:\n[1] https://github.com/openstack/nova/blob/848de1f712838d8ac34d6d79288b0f281b3afd19/nova/console/serial.py#L40","accounts_in_message":[],"_revision_number":6},{"id":"c292e3dd74c89aa07a01838cef1fe88b3f0a127a","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2016-03-09 15:33:06.000000000","message":"Patch Set 6: Code-Review+1\n\nI was thinking about that hard reboot your are talking on some comments above but we do not release ports, so we are good.","accounts_in_message":[],"_revision_number":6},{"id":"dae4ddabe1a012de99f17f72b8cf2b507b6083b3","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-11 09:34:41.000000000","message":"Topic changed from refactoring-libvirt to bug/1489853","accounts_in_message":[],"_revision_number":6},{"id":"8e19af2937b2b45e5909ae2052626c2678d2b07a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-03-14 09:05:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/275800/6/2887 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"437fc06b58d0bfbda70bf43c9b6a727c291d6dad","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-03-14 22:13:30.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"10b498703853fdf2ed0351ae76feeaa782baaf4b","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-15 12:23:54.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"99492a85f2889ac991230f7cbec67ddecfb74806","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-03-15 12:43:34.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/fe0014bf21fc460b8db6946706dadddd/index.html : SUCCESS in 15m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/614e72cbeb004790853b67abd8f0d2ce/index.html : SUCCESS in 16m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9082c1cca9e24d6bb72dae1414fc7e71/index.html : SUCCESS in 15m 33s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/7/check/real-db-upgrade_nova_percona_user_001:th-percona/64fa2371eacf49f1ba819da876a6fb91/index.html : SUCCESS in 18m 22s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/00/275800/7/check/real-db-upgrade_nova_mysql_user_002:th-mysql/554630584e114fd0b29f7bcdd64beba0/index.html : SUCCESS in 15m 22s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/00/275800/7/check/real-db-upgrade_nova_percona_user_002:th-percona/7b64671642c74dff8e0b4ccd138e62e0/index.html : SUCCESS in 15m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"4d201cdd73983606d6fb63865fcca2b0232f48d1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-03-15 12:50:36.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/275800/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"15cad0e0cad442efd117f4956316c1d99de23344","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-03-15 13:29:07.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/00/275800/7/check/check-ibm-tempest-dsvm-full/847a246/ : SUCCESS in 1h 04m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c81e25350c3fd278ba38f235e35c1df202c5e13a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-03-15 13:45:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/00/275800/7/check/dsvm-tempest-xen/a7dfdf0 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":7},{"id":"9897dfa1b0efae4ebfd88b9df89f7b44d74c97b2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-03-15 13:46:10.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/275800/7/9235/index.html : SUCCESS in 01 h 19 m 20 s","accounts_in_message":[],"_revision_number":7},{"id":"ab9ac5fcca2d8ef4a19038b320aefeca17731b85","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-03-15 14:04:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/275800/7/ : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":7},{"id":"ad00dfc9bb3ed5b86a64d1513381b5b83b789ed5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-15 14:32:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/00/275800/7/check/gate-nova-pep8/31408b8/ : SUCCESS in 2m 38s\n- gate-nova-docs http://docs-draft.openstack.org/00/275800/7/check/gate-nova-docs/d5c8d4c//doc/build/html/ : SUCCESS in 4m 16s\n- gate-nova-python27 http://logs.openstack.org/00/275800/7/check/gate-nova-python27/83fa58b/ : SUCCESS in 11m 41s\n- gate-nova-python34 http://logs.openstack.org/00/275800/7/check/gate-nova-python34/89444dd/ : SUCCESS in 31m 15s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-full/76afdbd/ : SUCCESS in 1h 09m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-postgres-full/5357c76/ : SUCCESS in 43m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-neutron-full/494db16/ : SUCCESS in 58m 35s\n- gate-grenade-dsvm http://logs.openstack.org/00/275800/7/check/gate-grenade-dsvm/5c69e8f/ : SUCCESS in 38m 58s\n- gate-nova-tox-db-functional http://logs.openstack.org/00/275800/7/check/gate-nova-tox-db-functional/992ff5a/ : SUCCESS in 12m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/00/275800/7/check/gate-grenade-dsvm-multinode/e36c260/ : SUCCESS in 54m 04s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/dc558fd/ : SUCCESS in 37m 47s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-cells/a514ca2/ : SUCCESS in 39m 40s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-full-ceph/129700c/ : SUCCESS in 29m 35s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/f74acfb/ : SUCCESS in 43m 48s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/00/275800/7/check/gate-tempest-dsvm-multinode-full/f305695/ : SUCCESS in 53m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2549c15529318b41c637c44e91355a5906d14b74","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-03-15 15:54:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/00/275800/7/check/check-dsvm-tempest-vz7-exe-minimal/d6f7610 : SUCCESS in 2h 04m 03s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"ab0702c8144893b47c5a79a53db0f8f9ed822c3c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-03-15 16:03:19.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-275800-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"690ec3613dc9ab6f70cdccc033c4389a1cbde591","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-03-15 17:05:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/275800/7/Nova-ML2-Sriov : SUCCESS in 43m 29s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"e4bebda77f5998acb812d7dfe33d05301b2572a7","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2016-03-31 14:48:39.000000000","message":"Patch Set 7: Code-Review-1\n\nAs we have discussed on IRC we should to not to try to use same ports but just release them","accounts_in_message":[],"_revision_number":7},{"id":"aaccf79f781482632fa167aec6397bfebaf22dc9","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-03-31 14:49:55.000000000","message":"Patch Set 7: Workflow-1\n\nWhen sahid pushed his changes this change here should be abandoned.","accounts_in_message":[],"_revision_number":7},{"id":"031141157fed784b8d1010d599bc04007efa67f1","author":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"date":"2016-04-06 13:53:15.000000000","message":"Abandoned\n\nSahid is solving bug 1489853 in change https://review.openstack.org/301247","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2f75119b43b49634688364ff8ad5465d691e044d","revisions":{"c64db1636bece461ddcc9a893fcb7b7eaf88ce72":{"kind":"REWORK","_number":1,"created":"2016-02-03 16:25:52.000000000","uploader":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"ref":"refs/changes/00/275800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/1"}}},"commit":{"parents":[{"commit":"11dfd9def3945cf2cf2dde4b1bc5d2b75cecbe3a","subject":"Merge \"Trivial: Remove a duplicated word\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/11dfd9def3945cf2cf2dde4b1bc5d2b75cecbe3a"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Markus Zoeller","email":"mzoeller@de.ibm.com","date":"2016-02-03 16:21:07.000000000","tz":60},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c64db1636bece461ddcc9a893fcb7b7eaf88ce72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c64db1636bece461ddcc9a893fcb7b7eaf88ce72"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"11dfd9def3945cf2cf2dde4b1bc5d2b75cecbe3a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5174166ae75dabdbe5eebb914bce4346a98bc491":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2016-02-04 16:12:07.000000000","uploader":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"ref":"refs/changes/00/275800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/2"}}},"commit":{"parents":[{"commit":"dcb63f0498c1e41081927ef81448c007b38cd8cd","subject":"Merge \"Remove prelude from disk-weight-sch reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dcb63f0498c1e41081927ef81448c007b38cd8cd"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Markus Zoeller","email":"mzoeller@de.ibm.com","date":"2016-02-04 13:48:12.000000000","tz":60},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists. This is also important when doing a live migration because\nthat code path calls \"_get_guest_config()\" multiple times.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5174166ae75dabdbe5eebb914bce4346a98bc491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5174166ae75dabdbe5eebb914bce4346a98bc491"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dcb63f0498c1e41081927ef81448c007b38cd8cd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9102f087d76089529ee00980af23246ab3444de8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-02-08 10:40:46.000000000","uploader":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"ref":"refs/changes/00/275800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/3"}}},"commit":{"parents":[{"commit":"5c8edcc6973191a44a83da19d8b8ee6486684809","subject":"Merge \"parametrize max_api_version in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c8edcc6973191a44a83da19d8b8ee6486684809"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Markus Zoeller","email":"mzoeller@de.ibm.com","date":"2016-02-08 09:55:48.000000000","tz":60},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists. This is also important when doing a live migration because\nthat code path calls \"_get_guest_config()\" multiple times.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9102f087d76089529ee00980af23246ab3444de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9102f087d76089529ee00980af23246ab3444de8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5c8edcc6973191a44a83da19d8b8ee6486684809","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"23ec0f82e11d7cdeaee31dad931cd8b3a4ac1d28":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-02-09 10:51:24.000000000","uploader":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"ref":"refs/changes/00/275800/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/4"}}},"commit":{"parents":[{"commit":"ae63dc98d516dd22b05dbce900d9d9b8f31f5394","subject":"Merge \"Add new DB API method to retrieve migration for instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae63dc98d516dd22b05dbce900d9d9b8f31f5394"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Markus Zoeller","email":"mzoeller@de.ibm.com","date":"2016-02-09 10:17:25.000000000","tz":60},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists. This is also important when doing a live migration because\nthat code path calls \"_get_guest_config()\" multiple times.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23ec0f82e11d7cdeaee31dad931cd8b3a4ac1d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23ec0f82e11d7cdeaee31dad931cd8b3a4ac1d28"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ae63dc98d516dd22b05dbce900d9d9b8f31f5394","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e6e537dd31bb27705d2d19c72f95dbbc1578c127":{"kind":"REWORK","_number":5,"created":"2016-02-23 12:52:45.000000000","uploader":{"_account_id":13816,"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","username":"pavone"},"ref":"refs/changes/00/275800/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/5"}}},"commit":{"parents":[{"commit":"9bd8ae378c1f69e7d973475e23fa510beaa9452b","subject":"Merge \"Improve the help text for cells options (6)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9bd8ae378c1f69e7d973475e23fa510beaa9452b"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Marco Pavone","email":"marco.pavone@de.ibm.com","date":"2016-02-23 08:27:06.000000000","tz":0},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists. This is also important when doing a live migration because\nthat code path calls \"_get_guest_config()\" multiple times.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\nCo-Authored-By: Marco Pavone \u003cmarco.pavone@de.ibm.com\u003e\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e6e537dd31bb27705d2d19c72f95dbbc1578c127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e6e537dd31bb27705d2d19c72f95dbbc1578c127"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9bd8ae378c1f69e7d973475e23fa510beaa9452b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1f65d64957951df91eec2c7a9202298f84102ecd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2016-03-08 09:35:29.000000000","uploader":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"ref":"refs/changes/00/275800/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/6"}}},"commit":{"parents":[{"commit":"6b2e8ed9bd7bd068b23879a60f634ac94cf563ec","subject":"Merge \"Fix string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b2e8ed9bd7bd068b23879a60f634ac94cf563ec"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Markus Zoeller (markus_z)","email":"mzoeller@de.ibm.com","date":"2016-03-08 09:35:28.000000000","tz":0},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists. This is also important when doing a live migration because\nthat code path calls \"_get_guest_config()\" multiple times.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\nCo-Authored-By: Marco Pavone \u003cmarco.pavone@de.ibm.com\u003e\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f65d64957951df91eec2c7a9202298f84102ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f65d64957951df91eec2c7a9202298f84102ecd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6b2e8ed9bd7bd068b23879a60f634ac94cf563ec","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2f75119b43b49634688364ff8ad5465d691e044d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2016-03-15 12:23:54.000000000","uploader":{"_account_id":11303,"name":"Markus Zoeller (markus_z)","email":"mzoeller@linux.vnet.ibm.com","username":"mzoeller"},"ref":"refs/changes/00/275800/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/275800/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/275800/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/275800/7"}}},"commit":{"parents":[{"commit":"aecefcb9dd5403a0d0a83e9f95d49dcea22e5fc4","subject":"Merge \"Add reno for disco driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aecefcb9dd5403a0d0a83e9f95d49dcea22e5fc4"}]}],"author":{"name":"Alexander Schmidt","email":"alexs@linux.vnet.ibm.com","date":"2016-02-03 15:47:32.000000000","tz":60},"committer":{"name":"Markus Zoeller (markus_z)","email":"mzoeller@de.ibm.com","date":"2016-03-15 12:23:53.000000000","tz":0},"subject":"libvirt: prevent serial console port exhaustion","message":"libvirt: prevent serial console port exhaustion\n\nWhen an instance gets rebuild, new serial console ports will be\ncreated although they could be reused. This leads to a port exhaustion.\n\nThis patch reuses the once created ports in case the instance already\nexists. This is also important when doing a live migration because\nthat code path calls \"_get_guest_config()\" multiple times.\n\nRelated-Bug: 1455252\nCloses-Bug: 1489853\nCo-Authored-By: Marco Pavone \u003cmarco.pavone@de.ibm.com\u003e\n\nChange-Id: I6d8f8542e517cd1e7d2ed1c2bbf7bcf16505127c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f75119b43b49634688364ff8ad5465d691e044d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f75119b43b49634688364ff8ad5465d691e044d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aecefcb9dd5403a0d0a83e9f95d49dcea22e5fc4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
