)]}'
{"id":"openstack%2Fnova~279626","triplet_id":"openstack%2Fnova~master~I93f0a2cc568b60c2b3f7509449167f03c3f30fb5","project":"openstack/nova","branch":"master","topic":"libvirt-instance-storage","hashtags":[],"change_id":"I93f0a2cc568b60c2b3f7509449167f03c3f30fb5","subject":"Rename Raw backend to Flat","status":"MERGED","created":"2016-02-12 16:07:09.000000000","updated":"2016-05-18 20:11:21.000000000","submitted":"2016-05-18 16:45:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92fc2cd8e8b1377fcaf3310c4cf0ed7ffbd794d7","_number":279626,"virtual_id_number":279626,"owner":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-05-18 16:45:33.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":0,"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},{"value":1,"date":"2016-05-17 12:17:17.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"date":"2016-05-18 20:11:21.000000000","post_submit":true,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2016-05-18 19:00:47.000000000","post_submit":true,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":2,"date":"2016-05-17 19:12:24.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":0,"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":2,"date":"2016-05-17 14:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-05-17 18:59:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":0,"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":1,"date":"2016-05-18 14:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"value":0,"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},{"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},{"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-12 17:50:33.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2016-02-21 09:51:36.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2016-02-23 10:38:39.000000000","updated_by":{"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"},"reviewer":{"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"},"state":"REVIEWER"},{"updated":"2016-04-12 15:31:03.000000000","updated_by":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"reviewer":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"state":"REVIEWER"},{"updated":"2016-04-26 09:32:39.000000000","updated_by":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"reviewer":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"state":"REVIEWER"},{"updated":"2016-04-30 19:55:39.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-05 19:42:02.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-05 21:32:35.000000000","updated_by":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"reviewer":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"state":"REVIEWER"},{"updated":"2016-05-06 15:18:12.000000000","updated_by":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"reviewer":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"state":"REVIEWER"},{"updated":"2016-05-06 16:01:28.000000000","updated_by":{"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},"reviewer":{"_account_id":7461,"name":"Paul Murray","email":"pmurray@hpe.com","username":"ptm"},"state":"REVIEWER"},{"updated":"2016-05-06 18:24:12.000000000","updated_by":{"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},"reviewer":{"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},"state":"REVIEWER"},{"updated":"2016-05-08 16:53:04.000000000","updated_by":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"reviewer":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"state":"REVIEWER"},{"updated":"2016-05-10 07:53:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-12 09:15:55.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-12 18:39:46.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-13 14:42:24.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-13 16:09:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-13 16:46:00.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-13 17:17:24.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-13 22:53:26.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-17 12:17:17.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-17 18:59:47.000000000","updated_by":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"reviewer":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"state":"REVIEWER"},{"updated":"2016-05-17 19:12:24.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2016-05-18 14:35:41.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2016-05-18 16:45:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-05-18 19:00:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-18 20:11:21.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"}],"messages":[{"id":"f93955ddef492199db80e51a467d08382904ead6","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-12 16:07:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b96f0fd48a8cdab10cc20f8ac08b6e071df5274","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-12 16:37:09.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/448d0e16f3bb4daf9baeb17ebe319cc2/index.html : SUCCESS in 14m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/226d5c08252b403ea1c049509865ff9e/index.html : SUCCESS in 15m 48s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c33bd157ee9149debd743e96b8ead97f/index.html : SUCCESS in 14m 35s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_percona_user_001:th-percona/8531e731d1074c2186f3edc01250db14/index.html : SUCCESS in 14m 32s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7ab2ab6d8256454e9ddf09f113b472bc/index.html : SUCCESS in 13m 43s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_percona_user_002:th-percona/741ba3d43aab4336bc980df5b9b81ef0/index.html : SUCCESS in 11m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"97230defb8b07d6a34047b9464c99780d44da2c5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-12 16:49:03.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/279626/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d1181a5af475976b88e0b27400711b1a1d4807d9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-12 17:43:18.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/1/6188/index.html : SUCCESS in 01 h 11 m 02 s","accounts_in_message":[],"_revision_number":1},{"id":"f78f0ca7acaf237fc5157a9e237c2f58d31b2777","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2016-02-12 17:50:33.000000000","message":"Patch Set 1: Code-Review+1\n\nNice, thanks for the clean up and clarifying comment in imagebackend.py. FWIW, makes it relatively easier to follow for those (like me) who\u0027re not deeply familiar with this area.","accounts_in_message":[],"_revision_number":1},{"id":"4dc312dfd9a68d525a0642ebd3abc7022aef4027","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 17:51:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/1/check/check-dsvm-tempest-vz7-exe-minimal/5fdf4a1 : SUCCESS in 49m 53s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"cd115d04ed6ee0059dfe684161bc0dce3fc3a29b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 17:54:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/1/check/gate-nova-pep8/0823a44/ : SUCCESS in 2m 30s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/1/check/gate-nova-docs/d2c9d9d//doc/build/html/ : SUCCESS in 8m 38s\n- gate-nova-python27 http://logs.openstack.org/26/279626/1/check/gate-nova-python27/9e33ea2/ : FAILURE in 9m 24s\n- gate-nova-python34 http://logs.openstack.org/26/279626/1/check/gate-nova-python34/b6e97fc/ : FAILURE in 26m 32s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-full/a0b7d31/ : FAILURE in 47m 40s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-postgres-full/c410288/ : SUCCESS in 36m 48s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-neutron-full/78da370/ : FAILURE in 47m 38s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/1/check/gate-grenade-dsvm/72938cc/ : SUCCESS in 32m 24s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/1/check/gate-nova-tox-functional/d9dd797/ : SUCCESS in 17m 26s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/1/check/gate-grenade-dsvm-multinode/c2ca8e3/ : SUCCESS in 43m 37s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/32d7a62/ : FAILURE in 24m 30s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-cells/96eec3b/ : SUCCESS in 37m 17s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-full-ceph/489faca/ : SUCCESS in 36m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/3516d07/ : SUCCESS in 40m 05s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-multinode-full/b3947c8/ : SUCCESS in 1h 04m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1598985cca16e8b1243def6a02cb7ef2da42d96","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-02-12 18:47:15.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27: unrecognized error\n- gate-nova-python34: unrecognized error\n- gate-tempest-dsvm-full: unrecognized error\n- gate-tempest-dsvm-neutron-full: https://bugs.launchpad.net/bugs/1357055\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"6a2d9d1cb38fd68c7ebea84c488128c6ca38e866","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-12 19:01:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/1/ : SUCCESS in 2h 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f85efe9a82be7da6daffbdab172b38dc5035e118","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2016-02-16 10:42:24.000000000","message":"Patch Set 1:\n\nrecheck bug 1357055","accounts_in_message":[],"_revision_number":1},{"id":"161ec0e1c109e86e1ae03a1beb91ee1a8c562617","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-16 11:38:22.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/26a7e9fee3514e48bfdb0b3f55dd5cfe/index.html : SUCCESS in 14m 24s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d8cd281fcfc24b7889fb1740ea3a7b3e/index.html : SUCCESS in 17m 18s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/9d698f30d5d04c78b87efff7309923a8/index.html : SUCCESS in 14m 41s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_percona_user_001:th-percona/740f7f68b425467ebdc8727557d4b0b9/index.html : SUCCESS in 14m 57s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_mysql_user_002:th-mysql/dc165454cd5246bebc9f92461fea7140/index.html : SUCCESS in 14m 00s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/1/check/real-db-upgrade_nova_percona_user_002:th-percona/cf709140ca244c1185f14ebebbc3cb0c/index.html : SUCCESS in 14m 16s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"03a2c3c9216b4fbfc62c49ef3f295b8af8a11cc6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-16 11:45:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/1/check/gate-nova-pep8/4038c52/ : SUCCESS in 3m 15s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/1/check/gate-nova-docs/c721ff3//doc/build/html/ : SUCCESS in 7m 15s\n- gate-nova-python27 http://logs.openstack.org/26/279626/1/check/gate-nova-python27/3b7ae78/ : FAILURE in 11m 35s\n- gate-nova-python34 http://logs.openstack.org/26/279626/1/check/gate-nova-python34/6c9fff2/ : FAILURE in 30m 10s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-full/63685c1/ : FAILURE in 37m 00s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-postgres-full/9e28df2/ : SUCCESS in 36m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-neutron-full/8d2b06c/ : FAILURE in 45m 18s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/1/check/gate-grenade-dsvm/b330c8e/ : SUCCESS in 32m 09s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/1/check/gate-nova-tox-functional/12468fd/ : SUCCESS in 12m 29s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/1/check/gate-grenade-dsvm-multinode/f0e7dbc/ : SUCCESS in 41m 40s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ffef2f0/ : SUCCESS in 53m 04s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-cells/137525c/ : SUCCESS in 44m 59s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-full-ceph/80506db/ : SUCCESS in 1h 02m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/3bacdfb/ : SUCCESS in 1h 00m 58s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/1/check/gate-tempest-dsvm-multinode-full/a044f37/ : SUCCESS in 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bbd57fe98c0922369f15757a8d130c6dde2dfd9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-16 12:55:47.000000000","message":"Patch Set 1:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/1 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"ddbe9bfd374ab09258cb2cf32c668958ab360862","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-16 12:57:07.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/1/ : FAILURE in 2h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"975c325a6d00dac120bda73e3f2c0b85798717d9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-17 15:12:49.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/279626/1/2310 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cd50ad4f8527b9c913b99cc9528ced4e5a2c3f04","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-19 17:19:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80b29abf5826c395e6d0df5c95ea491a4aaf72ef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-19 17:36:07.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"e6a895baf391563f97c8c2c27f92a770408cc8ca","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-19 17:36:24.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"c29f0b977a98a4f0dc4cb81d5e825d8687971561","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-19 17:36:27.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"a6a28e94d533123d4dbcf1e2a36e7dea68a521b0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-19 17:37:09.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/279626/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ef109fab59d74b7d39710dc95d59e87fec3fcbb5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-19 17:38:13.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f6638b5b3c4e4265afe8e02d8fb33c57/index.html : SUCCESS in 12m 23s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/f62f74f71d004da9a833362f8f74034e/index.html : SUCCESS in 16m 13s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a66beb75944c4d36b0a6012bb6895d5b/index.html : SUCCESS in 14m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/2/check/real-db-upgrade_nova_percona_user_001:th-percona/b0d7680d30a34b8294c9671f6876a1da/index.html : SUCCESS in 15m 46s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/2/check/real-db-upgrade_nova_mysql_user_002:th-mysql/333ab3a93d4e4db3a7ecd1cc6d24c9c8/index.html : SUCCESS in 13m 50s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/2/check/real-db-upgrade_nova_percona_user_002:th-percona/13bd7cc947ad431bb8f3e31d79aeac6d/index.html : SUCCESS in 14m 05s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"c3f38a397cc40082a4e011cd2265b487a9d4fbdd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-19 18:51:15.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/2/6758/index.html : SUCCESS in 01 h 27 m 28 s","accounts_in_message":[],"_revision_number":2},{"id":"bb74eb18b821bcaf96b2215a971cdaff6f505f1e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-19 23:39:56.000000000","message":"Patch Set 2:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/2 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"879f0485815c86f21937abfa2fa427a8d6a42275","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-20 00:25:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8c9c32d37bbcae235017ef9ea578741ac8ff6d86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-20 00:41:58.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"9c63c22543389842c75be97a7d987ef970b78fd2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-20 00:42:24.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"cf5697ef9d9a82d5421cac6be847f6855c24c6f7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-20 00:43:09.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/279626/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a79454667b3aa18b77070fe3be8b6f496098af2a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-20 00:43:44.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"958d48a1fc93f81ae56e992f01dd9c4200e3324d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-20 00:44:14.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/845549e082e348afba01f3e68696ca61/index.html : SUCCESS in 14m 33s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/89fc28d9dcd547ff9362e39ab4a1e9db/index.html : SUCCESS in 13m 22s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7f31426e827444eb9b751262c9a78cc5/index.html : SUCCESS in 14m 28s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/3/check/real-db-upgrade_nova_percona_user_001:th-percona/9ef6392eedea4d4fbf1c1cb60b25a029/index.html : SUCCESS in 15m 58s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/9eb64a0258e045a9a130ad8f30e09755/index.html : SUCCESS in 11m 55s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/3/check/real-db-upgrade_nova_percona_user_002:th-percona/4b7af62a8a3d459a8c59c2bada8072e9/index.html : SUCCESS in 14m 27s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"9be788294a59e4eb0b51abd59ebd9583b637dcec","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-20 01:37:01.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/3/6806/index.html : SUCCESS in 01 h 07 m 07 s","accounts_in_message":[],"_revision_number":3},{"id":"95a77c1a4d26b67bc7fe75bdc5c3b52d4b9fdbc7","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-20 22:30:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"83d4cc62eb707356e3594e4154ec9de2d6ca4540","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-20 23:33:53.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/3/check/check-ibm-tempest-dsvm-full/fa5380d/ : SUCCESS in 54m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3b01716915e0a019f2d43d0e2e1ae8f281947b6e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-20 23:56:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/3/check/gate-nova-pep8/1909840/ : SUCCESS in 3m 00s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/3/check/gate-nova-docs/04d8482//doc/build/html/ : SUCCESS in 7m 21s\n- gate-nova-python27 http://logs.openstack.org/26/279626/3/check/gate-nova-python27/76ec262/ : SUCCESS in 8m 22s\n- gate-nova-python34 http://logs.openstack.org/26/279626/3/check/gate-nova-python34/337385c/ : FAILURE in 40m 11s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full/56d932f/ : SUCCESS in 1h 07m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-postgres-full/7050855/ : SUCCESS in 36m 43s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-neutron-full/f1dfb9f/ : SUCCESS in 55m 39s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm/c167266/ : SUCCESS in 38m 34s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/3/check/gate-nova-tox-functional/c1e109e/ : SUCCESS in 12m 06s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm-multinode/f5ed25d/ : SUCCESS in 32m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0b0eb1d/ : SUCCESS in 34m 26s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-cells/ae5c098/ : SUCCESS in 34m 53s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-ceph/ea46b0e/ : SUCCESS in 1h 23m 37s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/8a2f7eb/ : SUCCESS in 54m 25s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-multinode-full/4d72f89/ : SUCCESS in 1h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d53b75920de0dc25517d98017b3317082525b53","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 02:06:15.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/3/ : FAILURE in 3h 30m 32s","accounts_in_message":[],"_revision_number":3},{"id":"fbddbd95a6e909850193d3cf7712bfb53fa8fa35","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 04:14:41.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"b31692562cd210ba58eb26756f8038bd6c95908d","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-21 08:54:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b75f8281805720c63d8809d22c77eb20af28dafc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-21 09:51:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/3/check/gate-nova-pep8/5b37f86/ : FAILURE in 30m 06s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/3/check/gate-nova-docs/62dc6dc//doc/build/html/ : SUCCESS in 7m 10s\n- gate-nova-python27 http://logs.openstack.org/26/279626/3/check/gate-nova-python27/2b14d7e/ : SUCCESS in 9m 57s\n- gate-nova-python34 http://logs.openstack.org/26/279626/3/check/gate-nova-python34/2919567/ : FAILURE in 40m 10s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full/e5fb211/ : SUCCESS in 53m 39s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-postgres-full/6bcc088/ : SUCCESS in 32m 56s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-neutron-full/3da98b2/ : SUCCESS in 56m 38s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm/a8ced19/ : SUCCESS in 32m 34s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/3/check/gate-nova-tox-functional/13455a4/ : SUCCESS in 12m 09s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm-multinode/ec283b3/ : SUCCESS in 40m 03s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/881eae2/ : SUCCESS in 38m 36s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-cells/858992f/ : SUCCESS in 28m 33s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-ceph/31fe50d/ : SUCCESS in 45m 32s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/ca7d520/ : SUCCESS in 34m 41s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-multinode-full/3e0f56f/ : FAILURE in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7de4db3d4fdd9834e4e2be146f40cd575f1d0634","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-02-21 09:51:36.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-pep8: https://bugs.launchpad.net/bugs/1542849\n- gate-nova-python34: https://bugs.launchpad.net/bugs/1542849\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"72ac2fc5607fb7467908ebe5005ab46f83096f64","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 09:52:11.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"401a7ea81c263b145b1c45bb49349e34fb01b783","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-21 09:56:55.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/3/check/check-ibm-tempest-dsvm-full/df5d52c/ : SUCCESS in 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f13e51cf9ea22219951662038274fc2f5bca39e7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 10:27:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/3/ : SUCCESS in 1h 29m 11s","accounts_in_message":[],"_revision_number":3},{"id":"3be20821657bc6bcc98191032f9b8388f9779bce","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-21 19:38:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"aeff2b441aac1a2f8348513a2eba8fc63e955f12","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 20:32:41.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"acdaf01b573793297298958326a9a69e69e9b1a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-21 20:37:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/3/check/gate-nova-pep8/467ce07/ : SUCCESS in 2m 29s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/3/check/gate-nova-docs/2a8d028//doc/build/html/ : SUCCESS in 6m 48s\n- gate-nova-python27 http://logs.openstack.org/26/279626/3/check/gate-nova-python27/f4ec73e/ : SUCCESS in 10m 01s\n- gate-nova-python34 http://logs.openstack.org/26/279626/3/check/gate-nova-python34/595fded/ : FAILURE in 29m 18s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full/0b4a9bb/ : SUCCESS in 30m 38s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-postgres-full/52e9a6e/ : SUCCESS in 35m 01s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-neutron-full/453ca57/ : SUCCESS in 50m 48s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm/1edcb8e/ : SUCCESS in 32m 14s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/3/check/gate-nova-tox-functional/3cdcb1c/ : SUCCESS in 13m 15s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm-multinode/083880a/ : SUCCESS in 42m 15s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6b81fad/ : SUCCESS in 29m 11s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-cells/9e4103e/ : SUCCESS in 36m 14s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-ceph/9e4ad05/ : SUCCESS in 48m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/257b119/ : SUCCESS in 57m 02s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-multinode-full/22c77f4/ : FAILURE in 50m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3298c35bc313733c3cc92d2501b187ac7350e7d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-21 20:41:02.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/3/check/check-ibm-tempest-dsvm-full/dff73a5/ : SUCCESS in 56m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8cd3b7e5a8c2a27f539cd37d87bb68d0bc8709b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 21:32:44.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/3/ : FAILURE in 1h 49m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b0fac96e395857aee64ec79b5f42a29de115b8f3","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-21 22:44:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"97d19083ffa69b482b8f273f58ea97bdecbbc448","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-21 23:35:26.000000000","message":"Patch Set 3:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/3 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"d75435d4a04e81b32b7d27bfd296ad851dfb9108","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-21 23:37:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/3/check/gate-nova-pep8/dcbbcf9/ : SUCCESS in 2m 22s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/3/check/gate-nova-docs/1e94afb//doc/build/html/ : SUCCESS in 6m 28s\n- gate-nova-python27 http://logs.openstack.org/26/279626/3/check/gate-nova-python27/abf2c3a/ : SUCCESS in 9m 31s\n- gate-nova-python34 http://logs.openstack.org/26/279626/3/check/gate-nova-python34/070484c/ : FAILURE in 27m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full/2eab3a3/ : SUCCESS in 41m 48s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-postgres-full/5e9314f/ : SUCCESS in 33m 31s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-neutron-full/edc863e/ : SUCCESS in 48m 38s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm/e051f70/ : SUCCESS in 31m 18s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/3/check/gate-nova-tox-functional/ad40a3b/ : SUCCESS in 11m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm-multinode/bd5311b/ : SUCCESS in 30m 42s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/44e9b88/ : SUCCESS in 27m 30s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-cells/087695c/ : SUCCESS in 36m 15s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-ceph/fd40dc0/ : SUCCESS in 34m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/9ab0c49/ : SUCCESS in 33m 27s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-multinode-full/8033bba/ : FAILURE in 50m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"913e55ad2769e4da0d52e8fdc0aab1c426fb8150","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-21 23:44:04.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/3/check/check-ibm-tempest-dsvm-full/4d84471/ : SUCCESS in 57m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba5523c390b124e917c42ae2ed59d31e2b267bdd","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-22 01:41:18.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/3/ : FAILURE in 2h 54m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b23e37c53556f1e4aa46917deb72efefff071cf1","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-22 09:57:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1bc0610ba98ea8d7a49adb5495e3903a01e57218","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-02-22 10:58:56.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/3/check/check-ibm-tempest-dsvm-full/3812ef3/ : SUCCESS in 56m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58c1b65d1478828d2afa5eb9739b0e405025a20c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-22 11:02:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-pep8 http://logs.openstack.org/26/279626/3/check/gate-nova-pep8/ea990e7/ : SUCCESS in 2m 36s\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/3/check/gate-nova-docs/490aad2//doc/build/html/ : SUCCESS in 6m 50s\n- gate-nova-python27 http://logs.openstack.org/26/279626/3/check/gate-nova-python27/6e3c3da/ : SUCCESS in 9m 34s\n- gate-nova-python34 http://logs.openstack.org/26/279626/3/check/gate-nova-python34/0016c81/ : FAILURE in 27m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full/2ef8f0e/ : SUCCESS in 32m 26s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-postgres-full/c8ad931/ : SUCCESS in 38m 26s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-neutron-full/66bfd01/ : SUCCESS in 57m 25s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm/a138fad/ : SUCCESS in 32m 15s\n- gate-nova-tox-functional http://logs.openstack.org/26/279626/3/check/gate-nova-tox-functional/13e7e1a/ : SUCCESS in 10m 51s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/3/check/gate-grenade-dsvm-multinode/9c5a9f0/ : SUCCESS in 45m 15s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9d0b680/ : SUCCESS in 29m 45s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-cells/1b57e9e/ : SUCCESS in 45m 42s\n- gate-tempest-dsvm-full-ceph http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-ceph/83780c2/ : SUCCESS in 33m 09s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/81c6618/ : SUCCESS in 56m 40s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/3/check/gate-tempest-dsvm-multinode-full/e8ef225/ : SUCCESS in 1h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8b00d42a3df4cceefe9dd1d4090d94a51c099a4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-02-22 13:30:36.000000000","message":"Patch Set 3:\n\nBuild Failed.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/3 : FAILURE\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":3},{"id":"9805c134986adbc7397bd06ba092ecf3f8d7a387","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-22 13:53:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/3/ : SUCCESS in 2h 41m 52s","accounts_in_message":[],"_revision_number":3},{"id":"8b3ccbb5f55a269a88b8347da9473dfc40ebfa3c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-23 07:02:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/279626/3/2480 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c44e1d4ec4a4c4d359f9b99f42fe5bb7820560f4","author":{"_account_id":20223,"name":"zhaolihui","email":"zhaolh@awcloud.com","username":"zhaolihui"},"date":"2016-02-23 10:38:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c6c4848d3d12976f52120953671b0ee7f505d8fb","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-23 14:05:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86416310801abfe0726ac17a02968ee99cdb2379","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-02-23 15:29:27.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"b928dd8c20aed8db1ab03dde73765227718adbb9","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-02-23 17:12:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6f6f7c4ebd2e5a53a4002dc8898b7b8938a8524a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-02-23 17:12:37.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"653a1fb6095a9de72e62ac0cd6379427e1644ce2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-02-23 17:13:00.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"5c90cf49bf129ee97f65a6cfb2c37b0e59aa4ff3","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-02-23 17:46:02.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/6299a4d0be8f42869c1caaecfa6ba134/index.html : SUCCESS in 15m 28s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/38c65ae8332d4eb5b4b4c8045d3f5aab/index.html : SUCCESS in 17m 23s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7d62436bad5943b0aed3ccdf79d931dd/index.html : SUCCESS in 15m 15s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/4/check/real-db-upgrade_nova_percona_user_001:th-percona/78339667fd804c3e9e59832eb7fa502c/index.html : SUCCESS in 16m 11s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/73ff2e9ffa1d43c092331fce6d56b73a/index.html : SUCCESS in 16m 04s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/4/check/real-db-upgrade_nova_percona_user_002:th-percona/ce8dc37c0d494178aa1daf1c50d822fd/index.html : SUCCESS in 14m 01s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"2f3d2caef27e2f78dd6434c61d8f34babb146e78","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-02-23 17:49:48.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/279626/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"24900dd17b6c47c2860b6247cf9cb496fea683f6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-02-23 18:05:05.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"3553e370e13dd2cafbdd850a907af1fa3b47924a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-23 18:05:13.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"97b8564b6e79b27796df7e0863ee5a4b0d92b915","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-02-23 18:37:51.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/4/7065/index.html : SUCCESS in 01 h 09 m 09 s","accounts_in_message":[],"_revision_number":4},{"id":"1961b44dd596030fc4de1355469003aabb8e5e77","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-02-24 09:38:59.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"1e155ab260d482e26e75657d4fb444ea7cc59661","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-02-25 13:08:26.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/279626/4/2563 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"99918101141ab701885582e4bae1a0e706cf43d5","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-04-07 13:02:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8c5c6f24aa168caaa333472d48d280bbf46a2354","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-07 13:32:38.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/c00dbf36ff3945ea86a11784e1129104/index.html : SUCCESS in 15m 24s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/302c39cb7e6144f9a08c9c5d61173121/index.html : SUCCESS in 15m 54s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1212a75848f04913a2f45301b079dd71/index.html : SUCCESS in 16m 13s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/5/check/real-db-upgrade_nova_percona_user_001:th-percona/83978f6f32974789b28af8fe82d02bd9/index.html : SUCCESS in 12m 50s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"45766c2d04905a8c1097834b3722b7e61f31fed1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-07 13:40:02.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/279626/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"d2b13490b928f9ef0a9c78d4255c0d696a97943d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-07 14:15:33.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/5/check/check-ibm-tempest-dsvm-full/ef10122/ : SUCCESS in 1h 08m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"feeb281696f5ea870367fc85ef22ae466f4b8c73","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-07 14:19:39.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/5/11524/index.html : SUCCESS in 01 h 12 m 13 s","accounts_in_message":[],"_revision_number":5},{"id":"fe5657b675ef21323a8eb3609a5fd9f49e30493c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-07 14:40:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/5/ : SUCCESS in 1h 35m 48s","accounts_in_message":[],"_revision_number":5},{"id":"b28b0fcc9eff196e45f69336b02bfe83ddbe7c84","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-07 15:03:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/5/check/gate-nova-docs/446d6bd//doc/build/html/ : SUCCESS in 5m 03s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/5/check/gate-nova-pep8/07c699b/ : SUCCESS in 2m 49s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/5/check/gate-nova-python27-db/64127f8/ : SUCCESS in 9m 12s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/5/check/gate-nova-python34-db/b17b957/ : FAILURE in 15m 24s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-full/e28fe78/ : SUCCESS in 1h 15m 40s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-postgres-full/16e7430/ : SUCCESS in 56m 51s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-neutron-full/b8372e1/ : SUCCESS in 1h 15m 15s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/5/check/gate-grenade-dsvm/2945bb7/ : SUCCESS in 27m 10s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/5/check/gate-nova-tox-db-functional/15c994d/ : SUCCESS in 12m 47s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/5/check/gate-grenade-dsvm-multinode/0776945/ : SUCCESS in 49m 49s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/393e731/ : SUCCESS in 23m 37s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-cells/743c84e/ : SUCCESS in 44m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/e421b28/ : SUCCESS in 38m 21s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/5/check/gate-tempest-dsvm-multinode-full/433dd55/ : SUCCESS in 1h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"81c5e38ce43e2f91f1a8012521cf482c1c3f054f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-07 15:11:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/5/check/dsvm-tempest-xen/987ccec : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":5},{"id":"06b256785867f20980afb9dc00a9ab5abab4e642","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-04-07 19:20:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/26/279626/5/check/dsvm-tempest-kvm/b31ca33 : SUCCESS in 47m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"3445f119a5f29f71b5d20eea0aee96750f11f386","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-08 00:07:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/5/check/check-dsvm-tempest-vz7-exe-minimal/95a684b : SUCCESS in 1h 15m 58s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"a6a9334a080de85804108a26e5cfa4af9056b255","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-04-08 10:26:47.000000000","message":"Patch Set 5:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/5 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":5},{"id":"b7762882b4df4092d42d76d07dc10937a80e4b01","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-09 18:09:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/279626/5/3743 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"cf446d4035326d6be10d4ec34e1bd249e18c7bfa","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-04-12 14:56:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b7caa401c96c2e4c84f0bb63ca6544a413aa28c5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-12 15:14:53.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/9083b008cfb94b8db96b02d0fe1a8c90/index.html : SUCCESS in 16m 30s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5f9d842e9e204037809766f15ef07045/index.html : SUCCESS in 12m 56s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/1bcc365c537644b2ae8b08fb6473214d/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/6/check/real-db-upgrade_nova_percona_user_001:th-percona/218db706b828409f9c269d9f03b5aa25/index.html : SUCCESS in 15m 46s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"c6523732426a5d63489cfabfc0f8b2996232ae4d","author":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"date":"2016-04-12 15:31:03.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nNice cleanup!","accounts_in_message":[],"_revision_number":6},{"id":"eb8f748e3bf542dbc23e27032fe2abdfc35e5614","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-12 16:04:24.000000000","message":"Patch Set 6: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/6/11998/index.html : SUCCESS in 01 h 06 m 17 s","accounts_in_message":[],"_revision_number":6},{"id":"a44b22050ba44d7f90481d834a4b912ca6890be8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-12 16:07:37.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/6/check/check-ibm-tempest-dsvm-full/279933b/ : SUCCESS in 1h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bda5bac9ddef5341ba9b55b278b25dac4d8b1217","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-12 16:28:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/6/check/gate-nova-docs/c64fa07//doc/build/html/ : SUCCESS in 2m 25s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/6/check/gate-nova-pep8/e0245ff/ : SUCCESS in 2m 34s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/6/check/gate-nova-python27-db/e6c82a9/ : SUCCESS in 14m 06s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/6/check/gate-nova-python34-db/0c9cb17/ : SUCCESS in 11m 26s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-full/067d0c9/ : SUCCESS in 49m 56s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-postgres-full/9d37863/ : SUCCESS in 44m 30s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-neutron-full/39954fb/ : SUCCESS in 1h 05m 07s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/6/check/gate-grenade-dsvm/3dcd807/ : SUCCESS in 23m 49s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/6/check/gate-nova-tox-db-functional/c8d37eb/ : SUCCESS in 14m 03s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/6/check/gate-grenade-dsvm-multinode/ed2a3f6/ : SUCCESS in 27m 16s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6bb3a51/ : SUCCESS in 53m 45s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-cells/3187941/ : SUCCESS in 32m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/a7731d6/ : SUCCESS in 34m 49s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/6/check/gate-tempest-dsvm-multinode-full/675e43b/ : FAILURE in 46m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a34307f99e438005bf69b05435cd62a0851d534b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-12 16:56:20.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/6/ : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":6},{"id":"a5edbc2cc8654b81cc176b9259687f0413d6ea6c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-04-12 17:41:21.000000000","message":"Patch Set 6:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/26/279626/6 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":6},{"id":"06312f788db031e8d02e6cb07885cd0b88431e55","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-12 17:47:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/6/check/dsvm-tempest-xen/91902c8 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":6},{"id":"b1cbcac9eb02dd7f105d3e7f2cc36200f6ff0a75","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-12 21:06:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/6/check/check-dsvm-tempest-vz7-exe-minimal/a596698 : SUCCESS in 1h 12m 39s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"c486581618bca7e24b8f9d056783927d98282fb1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-12 22:42:25.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-26-279626-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"784d0dcdfc77c528a4403fdb87c2ab3048583898","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-12 23:33:54.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/279626/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"1ee13aa6f02ece5fb4af00da59179d4d66f7dd14","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-13 16:07:46.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/279626/6/3887 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6c2b6f206fa4384001f7ef9222633d33a06d1604","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-04-13 22:33:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/6/check-nova/Nova-ML2-Sriov/c9b9b6e : FAILURE in 2h 31m 10s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/6/check-nova/Nova-MACVTAP-ML2-Sriov/8dcd0ae : SUCCESS in 1h 29m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"65d7fb4457b700a5fc8973b5ea151bdc2eb9247f","author":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"date":"2016-04-26 09:32:39.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f68e70cb4868db066854e4660fe336f47a79d86d","author":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"date":"2016-04-29 12:33:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"fb9bc96c81e3a9b0868e898e719922583f63a83e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-29 12:41:02.000000000","message":"Patch Set 7: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/7/13395/index.html : FAILURE in 05 m 09 s","accounts_in_message":[],"_revision_number":7},{"id":"bca27f3d6edc5ef7d82e1099978b008b6aba0bd8","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-29 12:46:57.000000000","message":"Patch Set 7: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/7/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3da25f6a0baf4726935697eef2d0ce25/index.html : SUCCESS in 10m 44s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/7/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/dcc945835d3e49c88549c6174f014efa/index.html : SUCCESS in 9m 49s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/7/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7883ce3dbd1146799f05d0a87edc7f7c/index.html : SUCCESS in 9m 12s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/7/check/real-db-upgrade_nova_percona_user_001:th-percona/3f4684bd44ac4e14b3e087e37cd03902/index.html : SUCCESS in 11m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":7},{"id":"9798539d592bcf427b8ebbf6222e272476c910da","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-29 12:47:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/7/check/dsvm-tempest-xen/4f8adf7 : FAILURE in 1m 48s","accounts_in_message":[],"_revision_number":7},{"id":"742576abfa4e9b5a7115468dd5bba05a88d54f01","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-29 12:53:48.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/7/check/check-ibm-tempest-dsvm-full/b0001f6/ : FAILURE in 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ebddd3d7bd4fcdcd9fb314386de44647e618595a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-29 12:58:12.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/279626/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"233e3e7df285a14c446f8236866ce5e59a80e62e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-29 13:48:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/7/check/gate-nova-docs/be718d1//doc/build/html/ : SUCCESS in 4m 08s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/7/check/gate-nova-pep8/497572a/ : SUCCESS in 6m 04s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/7/check/gate-nova-python27-db/a61a4fc/ : SUCCESS in 10m 50s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/7/check/gate-nova-python34-db/214639f/ : SUCCESS in 10m 58s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-full/0e863cc/ : SUCCESS in 32m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-postgres-full/c2c9d0c/ : SUCCESS in 34m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-neutron-full/9d4e75e/ : SUCCESS in 55m 16s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/7/check/gate-grenade-dsvm/38ea52a/ : SUCCESS in 32m 23s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/7/check/gate-nova-tox-db-functional/e5a5452/ : SUCCESS in 12m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/7/check/gate-grenade-dsvm-multinode/cba07b6/ : SUCCESS in 43m 02s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9184c00/ : SUCCESS in 23m 33s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-cells/e981a48/ : SUCCESS in 27m 45s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph/6c26ea5/ : SUCCESS in 39m 56s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-multinode-full/e3122be/ : FAILURE in 52m 14s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/7/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/65be593/ : SUCCESS in 1h 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bd8b690f4c11dfd8c08aeae835ff4f25e60459ee","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-04-30 15:33:27.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/7/check-nova/Nova-ML2-Sriov/b75d213 : FAILURE in 14m 47s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/7/check-nova/Nova-MACVTAP-ML2-Sriov/c6688e4 : SUCCESS in 45m 50s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"a864bdfd20c40a25ec6ab38baf10ce6223459792","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-30 19:55:39.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/279626/7/4363 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"e7db48b9e765825144ca4b42590cda7934a09683","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-03 13:15:31.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"e1b32f0d108dd38bd1fbf1a033f11a95973f8612","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-03 13:29:46.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/05fa79fb2cb84e6293c191dcbf3d2deb/index.html : SUCCESS in 9m 55s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/7141f03c549a4ba28c50a1ff2587513d/index.html : SUCCESS in 11m 37s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d78404b726234c02adbad454b2df7b62/index.html : SUCCESS in 10m 58s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/8/check/real-db-upgrade_nova_percona_user_001:th-percona/0ef9424c0c2b458a8d9d0385c83591b3/index.html : SUCCESS in 9m 53s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":8},{"id":"555f3fd295919678a68fa31725b0036ffebd9981","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-03 14:18:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/8/check/gate-nova-docs/128aaf0//doc/build/html/ : SUCCESS in 5m 08s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/8/check/gate-nova-pep8/f54c16b/ : SUCCESS in 6m 04s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/8/check/gate-nova-python27-db/1077749/ : SUCCESS in 10m 46s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/8/check/gate-nova-python34-db/e34990e/ : SUCCESS in 11m 20s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-full/d509a1f/ : SUCCESS in 46m 03s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-postgres-full/2825833/ : SUCCESS in 45m 27s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-neutron-full/7078eae/ : SUCCESS in 58m 26s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/8/check/gate-grenade-dsvm/0ce2f18/ : SUCCESS in 49m 02s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/8/check/gate-nova-tox-db-functional/e41eb17/ : SUCCESS in 12m 33s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/8/check/gate-grenade-dsvm-multinode/5caf8a0/ : SUCCESS in 42m 38s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/132861e/ : SUCCESS in 48m 39s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-cells/af4e0ac/ : SUCCESS in 32m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph/1c5a133/ : SUCCESS in 37m 29s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-multinode-full/31147d9/ : FAILURE in 51m 49s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/8/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/ea4294b/ : SUCCESS in 55m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4c753cc75f1ab52fa778fa0f0235d6395f840510","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-03 14:51:48.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/279626/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"aa5893646b8efa5619604c9c36249dd552ecb05c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-03 15:41:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/8/ : SUCCESS in 2h 24m 14s","accounts_in_message":[],"_revision_number":8},{"id":"77ef3c5b1e25896ea807ab1dc6facd8bf983bbc7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-03 16:11:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/8/check/dsvm-tempest-xen/b4de565 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":8},{"id":"efa491c930b4d52c8db858119d511eb1bc5e96d8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-05-03 16:15:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/26/279626/8/check/dsvm-tempest-kvm/0dd9475 : FAILURE in 1h 11m 47s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"f6a7c764020a5bbece53c7e4c10255fd5d1d06a9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-03 16:33:11.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/8/check/check-ibm-tempest-dsvm-full/c81acee/ : SUCCESS in 1h 02m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"67f53cc7ac9aec8fc9be8e398db6158b6490a361","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-03 18:40:33.000000000","message":"Patch Set 8: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/8/13697/index.html : SUCCESS in 01 h 09 m 32 s","accounts_in_message":[],"_revision_number":8},{"id":"016c45d0620c21846a35b76e7b461f4db858e7d7","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-05 14:24:07.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"293945df6c849b002a301d36c8ff631a8b791acd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-05 14:38:24.000000000","message":"Patch Set 9: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/9/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/d2e060e055c4488bb3555ba01d8ac4f8/index.html : SUCCESS in 9m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/9/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6088cc9893784219a162f707546d190f/index.html : SUCCESS in 12m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/9/check/real-db-upgrade_nova_mysql_user_001:th-mysql/d041a2ab21024f74ab315f0f569bc78b/index.html : SUCCESS in 11m 33s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/9/check/real-db-upgrade_nova_percona_user_001:th-percona/4c38806f2732449f9a9fab3986a56967/index.html : SUCCESS in 9m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":9},{"id":"f14402e2e05e1ae280dfc7e542c828c273068d0b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-05 15:41:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://logs.openstack.org/26/279626/9/check/gate-nova-docs/e14ba9e/ : FAILURE in 4m 18s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/9/check/gate-nova-pep8/1152144/ : SUCCESS in 6m 52s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/9/check/gate-nova-python27-db/625c2ab/ : SUCCESS in 11m 41s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/9/check/gate-nova-python34-db/b0034cd/ : SUCCESS in 9m 05s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-full/196f258/ : SUCCESS in 34m 16s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-postgres-full/98c7dd2/ : SUCCESS in 38m 59s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-neutron-full/a5c9d06/ : SUCCESS in 48m 12s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/9/check/gate-grenade-dsvm/0a4bca5/ : SUCCESS in 24m 42s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/9/check/gate-nova-tox-db-functional/be1a968/ : SUCCESS in 10m 43s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/9/check/gate-grenade-dsvm-multinode/ade1d11/ : SUCCESS in 41m 37s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/5c86e47/ : SUCCESS in 21m 55s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-cells/336793a/ : SUCCESS in 33m 42s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph/8836c66/ : SUCCESS in 40m 04s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-multinode-full/410399d/ : FAILURE in 1h 05m 15s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/baa3e85/ : SUCCESS in 1h 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"acd79722266edcad54dd28dbe6406191f3f542a1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-05 15:48:46.000000000","message":"Patch Set 9: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/9/13956/index.html : SUCCESS in 01 h 21 m 38 s","accounts_in_message":[],"_revision_number":9},{"id":"3adba59bfa9ba9274d7452bfe6b5e2312e9e3607","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-05 16:10:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/9/check/dsvm-tempest-xen/77de875 : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":9},{"id":"7d10688463c6aa554099119d5d4eee4f8dcb38f2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-05 17:48:57.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/9/check/check-ibm-tempest-dsvm-full/004783f/ : SUCCESS in 1h 03m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d2532f3632b8fe679c752b848e7524e8875ac083","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-05 17:49:31.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/9/ : FAILURE in 3h 09m 16s","accounts_in_message":[],"_revision_number":9},{"id":"061ef1067f3ad14b2a653c6975fd570ea986790c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-05 19:06:57.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/279626/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"996d01510c132f656737cc9c65563cdaa35f38d0","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-05-05 19:42:02.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/26/279626/9/check/dsvm-tempest-kvm/0d3079a : FAILURE in 1h 02m 18s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"d0a28e3a0614162a88f84e53102d403d9d06e4c9","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-05 19:48:42.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/9/check/check-dsvm-tempest-vz7-exe-minimal/aff662a : SUCCESS in 2h 18m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"273a77671707c69a92c2dcf8b9d1a9a2b2ccfd54","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-05-05 21:32:35.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"798f8fe400e795b7381db745211fdb6fdd1b8b1c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-05 22:46:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/9/check/gate-nova-docs/0e7ec14//doc/build/html/ : SUCCESS in 4m 57s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/9/check/gate-nova-pep8/124d6f1/ : SUCCESS in 7m 57s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/9/check/gate-nova-python27-db/aa02fb0/ : SUCCESS in 7m 59s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/9/check/gate-nova-python34-db/115c4c2/ : SUCCESS in 10m 29s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-full/66e77e4/ : SUCCESS in 46m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-postgres-full/863efa7/ : SUCCESS in 42m 24s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-neutron-full/670d727/ : SUCCESS in 1h 01m 10s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/9/check/gate-grenade-dsvm/78856a9/ : SUCCESS in 43m 10s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/9/check/gate-nova-tox-db-functional/43f3bd9/ : SUCCESS in 11m 36s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/9/check/gate-grenade-dsvm-multinode/b2d679c/ : SUCCESS in 42m 01s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/c040465/ : SUCCESS in 38m 44s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-cells/c48416c/ : FAILURE in 30m 25s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph/1ad99a2/ : SUCCESS in 53m 31s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-multinode-full/bc1b1e2/ : SUCCESS in 49m 46s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/649d1e8/ : SUCCESS in 56m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"226c9fa16a6503b8b7a4e67072846e4bceb58b82","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-05 23:53:10.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/9/check/check-ibm-tempest-dsvm-full/83d4970/ : SUCCESS in 1h 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7738fcf8f2e3d7960ff8f2517b82da0ede8954d5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-06 01:16:57.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/9/ : FAILURE in 2h 53m 18s","accounts_in_message":[],"_revision_number":9},{"id":"864a4157b43b894a36efe95c43975fd36afecbbe","author":{"_account_id":16907,"name":"Diana Clarke","email":"diana.joan.clarke@gmail.com","username":"diana-clarke"},"date":"2016-05-06 02:15:13.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"2cb1c3526190a444c1954f8c5641670eb2d007c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-06 03:21:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/9/check/gate-nova-docs/9062827//doc/build/html/ : SUCCESS in 4m 15s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/9/check/gate-nova-pep8/6fa3eae/ : SUCCESS in 6m 09s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/9/check/gate-nova-python27-db/cb319ae/ : SUCCESS in 15m 33s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/9/check/gate-nova-python34-db/3bf9509/ : SUCCESS in 10m 06s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-full/9852cb2/ : SUCCESS in 41m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-postgres-full/e7a26b4/ : SUCCESS in 38m 54s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-neutron-full/b1e9997/ : SUCCESS in 1h 00m 40s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/9/check/gate-grenade-dsvm/b26f7d3/ : SUCCESS in 38m 14s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/9/check/gate-nova-tox-db-functional/9d4d3b0/ : SUCCESS in 11m 25s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/9/check/gate-grenade-dsvm-multinode/781e6cd/ : SUCCESS in 41m 12s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/d0fb984/ : SUCCESS in 40m 20s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-cells/c8f4853/ : SUCCESS in 38m 34s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph/9754a78/ : SUCCESS in 39m 03s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-multinode-full/5f5ce19/ : SUCCESS in 58m 19s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/9/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/f706318/ : SUCCESS in 1h 00m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5396157fa70e1afe33724e35ec971e9d930e4705","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-06 03:26:07.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/9/check/check-ibm-tempest-dsvm-full/f9ed3ac/ : SUCCESS in 1h 04m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5d2ff3c3aaa1a59d83b0aa918a2a2d31835c038a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-06 05:38:50.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/9/ : FAILURE in 3h 22m 53s","accounts_in_message":[],"_revision_number":9},{"id":"cc375978bcf49401fec35d8917206c8de1ab9b9e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-06 06:14:34.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"9a3ebb825658ed868f912481dce15bbbf90d3268","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-06 12:21:55.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"04dce0e7892c1462f772232581b17f6bb439778e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-06 12:36:17.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0309306f934c499bb7f063c9c3bbc2de/index.html : SUCCESS in 11m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6944e1f9545a47aabef2125ea638ec47/index.html : SUCCESS in 12m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ce747df1c1c2408f83c15737b3359808/index.html : SUCCESS in 8m 53s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/10/check/real-db-upgrade_nova_percona_user_001:th-percona/04cf8d12b5414aff877ee046c24cbbfd/index.html : SUCCESS in 10m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"daf584832144320443884d0ad3ceec41d22b02a9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-06 13:00:28.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/279626/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"ff9f3a551fa8062163390a369c6ba99fe2422cf2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-06 13:34:50.000000000","message":"Patch Set 10: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/10/14084/index.html : SUCCESS in 01 h 09 m 41 s","accounts_in_message":[],"_revision_number":10},{"id":"140d06eea3bdbbc6937363565e5ceecace7abc92","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-06 13:41:12.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/10/check/check-ibm-tempest-dsvm-full/1109e19/ : FAILURE in 1h 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a00b8d8643c57ac08d6890e5ce444ef8f9fbcb48","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-06 13:48:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/10/check/dsvm-tempest-xen/b3e719b : SUCCESS in 1h 14m 01s","accounts_in_message":[],"_revision_number":10},{"id":"31bfda4d2f3b6431b52ea01f091f1f8ba35cc444","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-06 14:55:59.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/10/ : FAILURE in 2h 32m 39s","accounts_in_message":[],"_revision_number":10},{"id":"481631f92e060bcf27331b9d525fc59059ee40a9","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-05-06 15:18:12.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"bbc95699ebfeab49615f89a8faf000c9c7531af7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-06 16:57:37.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/10/check/check-dsvm-tempest-vz7-exe-minimal/5dfde7a : FAILURE in 37m 03s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"eaf58d82fe6a6c2792c03f88ba5503af855430d6","author":{"_account_id":18602,"name":"Gábor Antal","email":"antal@inf.u-szeged.hu","username":"antal"},"date":"2016-05-06 18:24:12.000000000","message":"Patch Set 10: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"bb5d0ac466a472c18196daba22dbee6fb85ca0a6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-08 02:58:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/10/check-nova/Nova-ML2-Sriov/72e31f3 : FAILURE in 46m 45s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/10/check-nova/Nova-MACVTAP-ML2-Sriov/21b1cb9 : SUCCESS in 46m 31s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"dc871ed7d1dd0999d163056f694f203c273bc551","author":{"_account_id":10453,"name":"octopuszhang","email":"zhangyf@rc.inesa.com","username":"1"},"date":"2016-05-08 16:53:04.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"772433c9a6e83d157bc39daebbc27831549119e8","author":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"date":"2016-05-09 14:37:22.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"bfb15f9c3da09678e71435e79ebecf70a1b55248","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-09 15:37:26.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/11/ : FAILURE in 25m 41s","accounts_in_message":[],"_revision_number":11},{"id":"dbf7c76ab2d2c12a1005e5e683032d6bbaa8a238","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-09 15:51:39.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/11/check/gate-nova-docs/a59b1a4//doc/build/html/ : SUCCESS in 5m 11s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/11/check/gate-nova-pep8/c8a2cd1/ : SUCCESS in 7m 04s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/11/check/gate-nova-python27-db/3494e6e/ : SUCCESS in 9m 05s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/11/check/gate-nova-python34-db/f2edbe7/ : SUCCESS in 7m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-full/ce622d6/ : SUCCESS in 43m 37s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-postgres-full/3c23d31/ : SUCCESS in 37m 59s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-neutron-full/8eee84c/ : SUCCESS in 56m 39s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/11/check/gate-grenade-dsvm/94809ea/ : SUCCESS in 38m 38s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/11/check/gate-nova-tox-db-functional/ac61314/ : SUCCESS in 13m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/11/check/gate-grenade-dsvm-multinode/3be800c/ : SUCCESS in 43m 30s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/748c433/ : FAILURE in 18m 30s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-cells/42fb691/ : SUCCESS in 37m 35s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-full-devstack-plugin-ceph/7c7b054/ : SUCCESS in 31m 49s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-multinode-full/04e3c0b/ : SUCCESS in 57m 46s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/11/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/3317d59/ : SUCCESS in 1h 06m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0813abd2b348e58e9a74c5a1542bb97e83b21aac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-09 15:59:15.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/11/check/check-ibm-tempest-dsvm-full/576d5f8/ : SUCCESS in 1h 07m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b1f7c31c86abbae74389813e134726a9cab22d39","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-09 16:19:27.000000000","message":"Patch Set 11: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/11/14301/index.html : SUCCESS in 01 h 16 m 33 s","accounts_in_message":[],"_revision_number":11},{"id":"7f76ebce5884fcc89d0e20d77bbd466a4e2f87cc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-09 16:22:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/11/check/dsvm-tempest-xen/7ac8576 : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":11},{"id":"a1cb4d000a3fd83f7b46d6e03febe77b31f8708f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-09 16:32:42.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/pci/279626/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"090e935eba7e3a9f29daeaade54e76e19d0f8b26","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-10 07:53:48.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/11/check-nova/Nova-ML2-Sriov/f6a036f : SUCCESS in 46m 47s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/11/check-nova/Nova-MACVTAP-ML2-Sriov/13c8a0f : SUCCESS in 46m 56s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"4533d1126420f28bf32baacfb9ec57069a9e0749","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-10 09:10:33.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8926d5e585a5463a9d7f5ce7aaa81a76/index.html : SUCCESS in 9m 59s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/708d725931d64a5fa757b3fd0c161ec9/index.html : SUCCESS in 10m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b6e4cb25b7eb4a61a1b509f2e252be8d/index.html : SUCCESS in 10m 36s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/11/check/real-db-upgrade_nova_percona_user_001:th-percona/ca801638a4c4406b88c1bca9cdae3f77/index.html : SUCCESS in 10m 47s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"13821ec9f396c038e9be9afce6564c9b0c105fa2","author":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"date":"2016-05-12 07:52:28.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"26fc92fdaab876d5524ac91fc629df509b2894fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-12 07:52:48.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"f6c0c02285af822e2d5ea57c17c24f52a9cda325","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-12 07:52:49.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"e90602c209849b963e94a81006ee319ff85eb7be","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-12 08:05:53.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/dfc43f642857403ab2b18a7c082ba4ef/index.html : SUCCESS in 9m 50s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/c6211a9a89194b04aa47f19f7d2e7685/index.html : SUCCESS in 10m 48s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b722b2d97060446189d08c5a55de9e12/index.html : SUCCESS in 8m 59s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/12/check/real-db-upgrade_nova_percona_user_001:th-percona/e98cd9626e6642f1a2620b02abfc4906/index.html : SUCCESS in 11m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"7140bf0da3a94ec03919f8f937279b9485689797","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-12 09:01:53.000000000","message":"Patch Set 12: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/12/14791/index.html : SUCCESS in 01 h 07 m 19 s","accounts_in_message":[],"_revision_number":12},{"id":"6f246035468aa9869b3e6e04886ae40a063c7765","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-12 09:12:23.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/12/check/check-ibm-tempest-dsvm-full/8f4c767/ : SUCCESS in 1h 03m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"50160231707c225cb1c834882b2f32912251c8d6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-12 09:15:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- 3rdparty-nfv-ci http://intel-openstack-ci-logs.ovh/26/279626/12/check/3rdparty-nfv-ci/5e250f2 : SUCCESS in 1h 20m 36s\n- 3rdparty-nfv-ci-intel-tests http://intel-openstack-ci-logs.ovh/26/279626/12/check/3rdparty-nfv-ci-intel-tests/003cda6 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":12},{"id":"768e570a4e386ff9a17d9cb1cfdf921230c68bf3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-12 09:20:59.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/12/ : FAILURE in 1h 27m 03s","accounts_in_message":[],"_revision_number":12},{"id":"675e7ff5432c2674973effdc20da1189ca4dfda0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-12 12:35:23.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/12/check/check-dsvm-tempest-vz7-exe-minimal/400ad2d : SUCCESS in 1h 11m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"bb2203833db6c2db6edb67a83893bd8920728e98","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-12 13:56:04.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"4fc746698b44283e9de5c4b41067408ab6eb904b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-12 15:17:24.000000000","message":"Patch Set 13: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/13/14875/index.html : SUCCESS in 01 h 15 m 58 s","accounts_in_message":[],"_revision_number":13},{"id":"18c950e12f0ff847a91c19d80cfdab1d97b4d28b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-12 15:55:20.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/13/check/gate-nova-docs/37459aa//doc/build/html/ : SUCCESS in 8m 51s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/13/check/gate-nova-pep8/42cfbae/ : SUCCESS in 6m 42s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/13/check/gate-nova-python27-db/f71005e/ : SUCCESS in 19m 37s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/13/check/gate-nova-python34-db/7fefb0a/ : SUCCESS in 11m 33s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-full/03b98e9/ : SUCCESS in 57m 40s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-postgres-full/bd04d7c/ : SUCCESS in 39m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-neutron-full/01c1e06/ : SUCCESS in 51m 59s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/13/check/gate-grenade-dsvm/2c8a966/ : SUCCESS in 40m 50s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/13/check/gate-nova-tox-db-functional/75b535f/ : SUCCESS in 12m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/13/check/gate-grenade-dsvm-multinode/bd03970/ : SUCCESS in 53m 27s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e4941b2/ : FAILURE in 40m 44s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-cells/436bf26/ : SUCCESS in 44m 33s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph/eba76b0/ : FAILURE in 3m 06s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-multinode-full/86b69c6/ : SUCCESS in 1h 00m 03s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/1daa73a/ : FAILURE in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"de6a9a34b3be4520a6065667fb75862ec3d720c4","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-12 16:08:52.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"115011d920296b5ccaeec196216f8bcb8be07b5a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-12 16:41:31.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/279626/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"5e1815ae7f53e20b536b7534ac138cafb50e9e8c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-12 17:23:02.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/13/check/check-ibm-tempest-dsvm-full/88ff354/ : SUCCESS in 1h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"542dc6ec917026040e23e6dfdbf9a889d4f7fc0a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-12 17:55:59.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/13/check/gate-nova-docs/908f2d5//doc/build/html/ : SUCCESS in 7m 00s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/13/check/gate-nova-pep8/12088ae/ : SUCCESS in 6m 44s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/13/check/gate-nova-python27-db/61c3b49/ : SUCCESS in 10m 32s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/13/check/gate-nova-python34-db/14de40f/ : SUCCESS in 11m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-full/8890e37/ : SUCCESS in 50m 31s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-postgres-full/b2e0b2e/ : SUCCESS in 47m 07s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-neutron-full/4eb1edd/ : SUCCESS in 1h 05m 00s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/13/check/gate-grenade-dsvm/96c93ba/ : SUCCESS in 34m 19s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/13/check/gate-nova-tox-db-functional/9bf7a3e/ : SUCCESS in 13m 53s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/13/check/gate-grenade-dsvm-multinode/64393fb/ : SUCCESS in 44m 12s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/874f562/ : FAILURE in 39m 58s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-cells/8e8f795/ : SUCCESS in 43m 07s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph/72a1efe/ : SUCCESS in 53m 42s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-multinode-full/1bbc0d6/ : SUCCESS in 56m 15s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/13/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/9dc97e8/ : FAILURE in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"25318fa86b89d3fddc4313f0919a67ff0af6139a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-12 18:39:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/279626/13/check/dsvm-tempest-xen/b50a93e : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":13},{"id":"68f32e3747218cb6a8ebadeac6d78542f205fe16","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-12 20:00:15.000000000","message":"Patch Set 13: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"d0236b24ec34f89ec20675fe9000218b5173edd1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-13 02:28:14.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/13/check/check-dsvm-tempest-vz7-exe-minimal/4430b2c : FAILURE in 1h 18m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"bf9e55c799022a8397450af46d7116e4d07cb35f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-13 04:23:10.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/279626/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"32fcda11ab7ca7c72d36d1e084734d0933aa69a1","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-13 14:28:52.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d3dbe2a908fe7ee2c892593313f78bd89268c4df","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-13 14:42:24.000000000","message":"Patch Set 14: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/14/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b1ed0fe10e1544e4adda527d8bde00d8/index.html : SUCCESS in 9m 00s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/14/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/cf61ca3ec8694803b395563731ca3a9a/index.html : SUCCESS in 10m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/26/279626/14/check/real-db-upgrade_nova_mysql_user_001:th-mysql/390b16ca81624bb6930c544cb4834bd6/index.html : SUCCESS in 9m 24s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/26/279626/14/check/real-db-upgrade_nova_percona_user_001:th-percona/8bf14febd73249bba7eba786bcb3ae5a/index.html : SUCCESS in 11m 26s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":14},{"id":"1082c9577e1298a5d8167acfbe3403ca49de5e9c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-13 15:47:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/14/check/gate-nova-docs/ee5eea9//doc/build/html/ : SUCCESS in 5m 41s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/14/check/gate-nova-pep8/24afb47/ : SUCCESS in 5m 39s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/14/check/gate-nova-python27-db/dead045/ : SUCCESS in 11m 23s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/14/check/gate-nova-python34-db/1e39073/ : SUCCESS in 10m 26s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-full/d929b40/ : FAILURE in 1h 00m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-postgres-full/4697fad/ : FAILURE in 38m 26s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-neutron-full/ed77a8c/ : SUCCESS in 46m 32s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/14/check/gate-grenade-dsvm/18033de/ : SUCCESS in 31m 17s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/14/check/gate-nova-tox-db-functional/d0d8bc9/ : SUCCESS in 12m 15s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/14/check/gate-grenade-dsvm-multinode/4544bce/ : SUCCESS in 53m 16s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e4efe5d/ : SUCCESS in 31m 29s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-cells/dfd9944/ : SUCCESS in 29m 09s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph/4eff3ea/ : SUCCESS in 41m 07s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-multinode-full/c0543b4/ : SUCCESS in 54m 49s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/2e0d235/ : SUCCESS in 53m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"93110036c061d21c86e07eb8e2b822e1c2c41941","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-13 15:47:08.000000000","message":"Patch Set 14: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/14/15037/index.html : SUCCESS in 01 h 15 m 07 s","accounts_in_message":[],"_revision_number":14},{"id":"c3308b61b089932f14790be3f20c54d3b4420dd0","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-13 15:52:35.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"72e1f0ace54a5f224a340c01d217e47e8c5f1a0c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-13 16:09:09.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/14/check/check-ibm-tempest-dsvm-full/b4e320b/ : SUCCESS in 1h 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1538e1c720225f0174a59094644dc01103f9fafd","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-13 16:46:00.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/14/check/check-dsvm-tempest-vz7-exe-minimal/b479c29 : FAILURE in 2h 07m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"0e1c6d4f6ee8334ebec242ea270f3afde798ec8f","author":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"date":"2016-05-13 16:58:50.000000000","message":"Patch Set 14: Code-Review-1\n\nIs NoBacking a good name for that class? It does not sufficiently explain the image backend nature for me.","accounts_in_message":[],"_revision_number":14},{"id":"165e0490d4a153c78700cf0b3336988b4bd69e9d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-13 17:17:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/14/ : SUCCESS in 2h 46m 47s","accounts_in_message":[],"_revision_number":14},{"id":"26e095d272ba44d4b2effc644b6caa7d0e47426e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-13 18:11:41.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/14/check/gate-nova-docs/c5044b1//doc/build/html/ : SUCCESS in 5m 02s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/14/check/gate-nova-pep8/389460a/ : SUCCESS in 8m 45s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/14/check/gate-nova-python27-db/38e069b/ : SUCCESS in 9m 13s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/14/check/gate-nova-python34-db/4d12e93/ : SUCCESS in 11m 58s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-full/eaa7444/ : SUCCESS in 47m 33s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-postgres-full/ebdfb5b/ : FAILURE in 54m 27s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-neutron-full/dd17ed6/ : SUCCESS in 1h 00m 34s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/14/check/gate-grenade-dsvm/2ddf772/ : SUCCESS in 46m 50s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/14/check/gate-nova-tox-db-functional/0bf31d3/ : SUCCESS in 9m 53s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/14/check/gate-grenade-dsvm-multinode/0e6bb48/ : SUCCESS in 43m 42s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/51c295d/ : SUCCESS in 43m 42s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-cells/dfa441f/ : SUCCESS in 32m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph/43b3d1a/ : SUCCESS in 45m 17s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-multinode-full/416096e/ : SUCCESS in 56m 10s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/14/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/3c2eace/ : SUCCESS in 1h 14m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0b78955f4ac36f00fae35b70414ec47433003cb5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-13 22:53:26.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/279626/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"b9ca30feda366c156b76c60293f4a80aa2313e44","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-05-15 04:27:16.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-26-279626-13 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":13},{"id":"506117d6b06e3b110fba335eacf72aa61a8b99e7","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-16 08:45:23.000000000","message":"Patch Set 14:\n\n@Feodor, I\u0027m not wedded to it and would use a better name if you have one. Got any suggestions?\n\nTBH, my principal goal is to replace it with something which isn\u0027t actively misleading. I\u0027d be happy calling it Frederick, because then at least you\u0027d have to go look what it does, and wouldn\u0027t assume that it always writes files in raw format.\n\nFor the moment, I think NoBacking is better than both Raw and Frederick :)","accounts_in_message":[],"_revision_number":14},{"id":"75aa7513eb0e29fbafafe0387c92560bbbaa4e0f","author":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"date":"2016-05-16 10:00:02.000000000","message":"Patch Set 14:\n\nMy concern is not about developers only, but about users as well. Because the class name goes to Nova parameters as is. And users do not like even to read doc, what to say about code.\n\nI could propose SingleFile (but ploop is a single file too, as no backing), RawOrSingleQcow2 (probably too long), RawOrNoBackingQcow2 (or ...NoBacked...?) Or even misterious Qcow2Raw.","accounts_in_message":[],"_revision_number":14},{"id":"4651f44183bb3597855db7d6c32382cc7551f236","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-16 15:31:44.000000000","message":"Patch Set 14:\n\nI fear this could descend into bikeshedding :) To me, NoBacking is a good enough name for a class whose defining behaviour is that the objects it creates do not have a backing file, regardless of file format. In the absence of something astoundingly better I say lets go with it and move on.","accounts_in_message":[],"_revision_number":14},{"id":"612647c1eb80080e6794c74145bc74c0c90124f4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-05-16 17:20:35.000000000","message":"Patch Set 14: Code-Review-1\n\nI also don\u0027t want to bikeshed on this, but it took me a while to figure out why NoBacking makes sense. I\u0027m not sure \"nobacking\" is really going to be a useful improvement over \"raw\" in the config, which is the bigger sign to me that maybe it\u0027s not right, more so than just the class name. Since we\u0027re introducing a new alias for this behavior in config, I think it\u0027s worth spending some effort on making sure we\u0027ve improved the naming and not just traded one non-obvious word for another.\n\nWhat about using the word \"Flat\"? It\u0027s pretty common parlance to refer to squashing a CoW file with its backing image as \"flattening\".","accounts_in_message":[],"_revision_number":14},{"id":"ebf224a3e90cfdf34489bec675ce6c0c2e6fab9f","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-17 09:36:01.000000000","message":"Patch Set 14:\n\nOk, I\u0027ll give you that :) I like Flat a lot better than NoBacking. It also has the advantage of being concise. I\u0027ll update it.","accounts_in_message":[],"_revision_number":14},{"id":"fcfb13e7143aa98560e0865e3ac48145b4927887","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-17 11:03:59.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"39f75979b1d81f93f105cb6ce9ba59e498972070","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-17 11:38:48.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/279626/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"2d1dd15186a450b4c1f521aae9476b1b30a42d9b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-17 12:17:17.000000000","message":"Patch Set 15: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/279626/15/15369/index.html : SUCCESS in 01 h 10 m 27 s","accounts_in_message":[],"_revision_number":15},{"id":"ff96989ae7fdfc2aaa9b6b7d69bc954e78033552","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-17 12:31:12.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/15/check/check-ibm-tempest-dsvm-full/a2da564/ : SUCCESS in 1h 22m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4cf9d05a810a4b556f53208f23e8ed81a72afc33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-17 13:34:11.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/15/check/gate-nova-docs/9914e6e//doc/build/html/ : SUCCESS in 6m 04s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/15/check/gate-nova-pep8/f27af38/ : SUCCESS in 7m 18s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/15/check/gate-nova-python27-db/0b0d6ac/ : SUCCESS in 10m 26s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/15/check/gate-nova-python34-db/fa61391/ : SUCCESS in 12m 04s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full/c55b1e2/ : SUCCESS in 41m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-postgres-full/3ea5af9/ : SUCCESS in 47m 35s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-full/4b6a07c/ : SUCCESS in 1h 09m 49s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm/15d0cc2/ : SUCCESS in 39m 12s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/15/check/gate-nova-tox-db-functional/31658e0/ : SUCCESS in 12m 02s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm-multinode/75d845b/ : SUCCESS in 42m 53s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/87a43ec/ : SUCCESS in 40m 45s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-cells/7d3c353/ : FAILURE in 37m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph/b160d3c/ : SUCCESS in 36m 24s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-multinode-full/c8731c0/ : SUCCESS in 58m 51s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/dd3952a/ : SUCCESS in 1h 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"954f8275e48dfad2554cdf2f4c7be4567579d7da","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-17 13:44:01.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/279626/15/check/check-dsvm-tempest-vz7-exe-minimal/f3d4b40 : SUCCESS in 1h 03m 20s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"b5b3c3fd38700a04c0dc5f5941d5f633b2d34503","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-05-17 14:04:05.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)\n\nrecheck keystone tests failed","accounts_in_message":[],"_revision_number":15},{"id":"6f1f9dd7bcd03c74b79d9e70c6bff03e01412b4f","author":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"date":"2016-05-17 18:59:47.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"80540c85893ed82b414de5e255b268ab7703cfc8","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-05-17 19:12:24.000000000","message":"Patch Set 15: Code-Review+2\n\nA config help text update would be nice as Dan points out.","accounts_in_message":[],"_revision_number":15},{"id":"b7b226d4753b64cf5552a4e89f91105b2a5b9c04","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-17 19:34:47.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/26/279626/15/check/tempest-dsvm-full-nfv/cd3a851 : FAILURE in 1h 42m 17s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/26/279626/15/check/tempest-dsvm-intel-nfv/2e443f9 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":15},{"id":"7cead4a64a6a5da86f9e0d8f447a4c940bb3cd8b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-17 23:08:11.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/15/check-nova/Nova-ML2-Sriov/bfc1d31 : SUCCESS in 45m 45s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/26/279626/15/check-nova/Nova-MACVTAP-ML2-Sriov/f612a87 : SUCCESS in 46m 07s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"72598ab6c1a35f98cd8bf7226d4d2b4e73e30752","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-18 08:29:57.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d4d7a443a35c2b90e66b4ddcea8bc9ac244f8b2c","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-18 08:30:21.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"60eb1e329958cb9da84abd077a3954b3242fd71b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-18 09:48:35.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/15/check/gate-nova-docs/1ef350b//doc/build/html/ : SUCCESS in 4m 21s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/15/check/gate-nova-pep8/7059e05/ : SUCCESS in 5m 41s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/15/check/gate-nova-python27-db/132a125/ : SUCCESS in 12m 29s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/15/check/gate-nova-python34-db/33ca143/ : SUCCESS in 11m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full/3ad8cd4/ : SUCCESS in 42m 49s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-postgres-full/9531294/ : FAILURE in 42m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-full/db9cede/ : SUCCESS in 1h 03m 40s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm/ad68f73/ : SUCCESS in 24m 12s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/15/check/gate-nova-tox-db-functional/011f0d3/ : SUCCESS in 10m 30s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm-multinode/eee925f/ : SUCCESS in 48m 50s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ad43bde/ : SUCCESS in 22m 54s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-cells/fb0d1fa/ : SUCCESS in 26m 36s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph/a3a56bc/ : SUCCESS in 35m 38s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-multinode-full/e9dc1f4/ : FAILURE in 59m 10s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/2fbae9d/ : SUCCESS in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3b2aa408da6b7b79d1b88f5256b2d278990972b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-18 10:05:39.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/15/check/check-ibm-tempest-dsvm-full/72b1775/ : SUCCESS in 1h 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6a5b06d2db88fb0a24315f369b3dec5dabb44e25","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-18 11:08:21.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/15/ : SUCCESS in 2h 37m 41s","accounts_in_message":[],"_revision_number":15},{"id":"f7eb73297a75951792d22a7708994af17af6e3e2","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-18 11:09:54.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"df9786c3741ae802fcf45a8bd0b6d6fb0f660754","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-18 12:33:38.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/15/check/check-ibm-tempest-dsvm-full/3b79ab9/ : SUCCESS in 1h 22m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"afd8f241d76951ecf25206ab89f0fd98f9c91f97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-18 12:38:51.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/15/check/gate-nova-docs/8e9c1be//doc/build/html/ : SUCCESS in 5m 30s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/15/check/gate-nova-pep8/297dcbd/ : SUCCESS in 5m 21s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/15/check/gate-nova-python27-db/c07006a/ : SUCCESS in 13m 03s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/15/check/gate-nova-python34-db/1226631/ : SUCCESS in 11m 49s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full/3086231/ : SUCCESS in 54m 29s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-postgres-full/5210db2/ : SUCCESS in 53m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-full/4035c76/ : SUCCESS in 59m 14s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm/16a43e7/ : SUCCESS in 43m 55s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/15/check/gate-nova-tox-db-functional/0b0bf8a/ : SUCCESS in 11m 10s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm-multinode/0a035ed/ : SUCCESS in 43m 46s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/e147ec2/ : SUCCESS in 40m 17s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-cells/d28df90/ : SUCCESS in 27m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph/5f80bed/ : FAILURE in 1h 03m 50s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-multinode-full/41be125/ : SUCCESS in 57m 39s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/85c2fcb/ : SUCCESS in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"04f454306f2013cb60d587525544b965093a2b1c","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2016-05-18 12:47:56.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"11256e4fecdc7b674ad4be68f65d1323a812c4e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-18 14:16:17.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/279626/15/check/check-ibm-tempest-dsvm-full/0ef1bf6/ : SUCCESS in 1h 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6ee1de0c5c0128097a538abcccf28de43bf49057","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-05-18 14:35:34.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nrecheck keystone fail uhgain","accounts_in_message":[],"_revision_number":15},{"id":"22f643ae23918d11d31b5d249e95b28b64d8e73c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-05-18 14:35:41.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"8ed59f61d2050dc49ce37b6efa175b3b92512cba","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-18 14:40:17.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/279626/15/ : FAILURE in 3h 30m 01s","accounts_in_message":[],"_revision_number":15},{"id":"a1b549c62a594a814c0b59835261342564ce0663","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-18 15:28:57.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/15/check/gate-nova-docs/9a50d39//doc/build/html/ : SUCCESS in 4m 38s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/15/check/gate-nova-pep8/a66c93b/ : SUCCESS in 7m 14s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/15/check/gate-nova-python27-db/d997b09/ : SUCCESS in 10m 51s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/15/check/gate-nova-python34-db/0b3739a/ : SUCCESS in 11m 01s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full/dd83abf/ : SUCCESS in 36m 57s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-postgres-full/9f26704/ : SUCCESS in 44m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-full/aba86d6/ : SUCCESS in 51m 24s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm/17fbb22/ : SUCCESS in 38m 25s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/15/check/gate-nova-tox-db-functional/e0612ba/ : SUCCESS in 13m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/15/check/gate-grenade-dsvm-multinode/178a96b/ : SUCCESS in 51m 37s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a37460c/ : SUCCESS in 46m 52s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-cells/eca5914/ : SUCCESS in 37m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph/34295db/ : SUCCESS in 45m 51s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-multinode-full/6400a42/ : SUCCESS in 1h 00m 56s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/26/279626/15/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/83cac00/ : SUCCESS in 59m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a2eb9583984a063afa038af642e2cad5f053e145","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-18 15:29:24.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":15},{"id":"c651f5f9a2348c22400822403a2947e3757caa04","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-18 16:05:02.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/26/279626/15/check/tempest-dsvm-full-nfv/7d987cd : FAILURE in 1h 42m 23s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/26/279626/15/check/tempest-dsvm-intel-nfv/2b69760 : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":15},{"id":"4fdb6a7caa6afab73d1ecc6600def765f05fe512","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-18 16:45:33.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/26/279626/15/gate/gate-nova-docs/87d74b9//doc/build/html/ : SUCCESS in 4m 10s\n- gate-nova-pep8 http://logs.openstack.org/26/279626/15/gate/gate-nova-pep8/7a84769/ : SUCCESS in 5m 59s\n- gate-nova-python27-db http://logs.openstack.org/26/279626/15/gate/gate-nova-python27-db/c8516b0/ : SUCCESS in 14m 43s\n- gate-nova-python34-db http://logs.openstack.org/26/279626/15/gate/gate-nova-python34-db/6e7475e/ : SUCCESS in 13m 49s\n- gate-tempest-dsvm-full http://logs.openstack.org/26/279626/15/gate/gate-tempest-dsvm-full/cca26cd/ : SUCCESS in 41m 57s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/26/279626/15/gate/gate-tempest-dsvm-postgres-full/34c324b/ : SUCCESS in 46m 00s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/26/279626/15/gate/gate-tempest-dsvm-neutron-full/8ea521b/ : SUCCESS in 1h 14m 07s\n- gate-grenade-dsvm http://logs.openstack.org/26/279626/15/gate/gate-grenade-dsvm/ba25ff6/ : SUCCESS in 38m 19s\n- gate-nova-tox-db-functional http://logs.openstack.org/26/279626/15/gate/gate-nova-tox-db-functional/1899cb6/ : SUCCESS in 13m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/26/279626/15/gate/gate-grenade-dsvm-multinode/fa6a278/ : SUCCESS in 46m 03s\n- gate-tempest-dsvm-cells http://logs.openstack.org/26/279626/15/gate/gate-tempest-dsvm-cells/969f6c8/ : SUCCESS in 35m 36s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/279626/15/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/1f910f8/ : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":15},{"id":"4a6ac72e83fd3322178c82145390f8db559a878f","date":"2016-05-18 16:45:34.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":15},{"id":"45fc9132cf4ef50685118cb8a7bbe38fd406d13f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-05-18 19:00:47.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-26-279626-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"231db081f25eed082d2435c20f5701282a6d8f96","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-05-18 20:11:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"cff4d78a4a7ddbb18cb875c66b3c56f04a6caea3","revisions":{"74eae0d18e2156994c36aca4d98501380a7ba8f1":{"kind":"REWORK","_number":1,"created":"2016-02-12 16:07:09.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/1"}}},"commit":{"parents":[{"commit":"98d0887dd08170fde9ba08bdc85475ca4fc7f10e","subject":"Make import_file abstract","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/98d0887dd08170fde9ba08bdc85475ca4fc7f10e"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-12 15:02:26.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-12 15:02:26.000000000","tz":0},"subject":"Rename Raw backend to Unbacked","message":"Rename Raw backend to Unbacked\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027Unbacked\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74eae0d18e2156994c36aca4d98501380a7ba8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74eae0d18e2156994c36aca4d98501380a7ba8f1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"98d0887dd08170fde9ba08bdc85475ca4fc7f10e","is_merged_in_target_branch":false,"change_id":"Iab65bab5d114f1069a03d6a22e28dfe8b7151048","change_number":278013,"patch_set_number":4,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"bd619344c26209d364ba850a682892cf4bd2aaf1":{"kind":"REWORK","_number":2,"created":"2016-02-19 17:19:25.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/2"}}},"commit":{"parents":[{"commit":"a4de2b986db8c08e98f8f425f5ac2cebd71b9520","subject":"Add a lock() context manager to image backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4de2b986db8c08e98f8f425f5ac2cebd71b9520"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-12 15:02:26.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-19 15:50:22.000000000","tz":0},"subject":"Rename Raw backend to Unbacked","message":"Rename Raw backend to Unbacked\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027Unbacked\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd619344c26209d364ba850a682892cf4bd2aaf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd619344c26209d364ba850a682892cf4bd2aaf1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a4de2b986db8c08e98f8f425f5ac2cebd71b9520","is_merged_in_target_branch":false,"change_id":"Id9d469ac56ad61ff8273179c0edbf8a31f716e89","change_number":279625,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"1f8a60096202973edbc27a2c8bcc191d65e60a37":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-02-20 00:25:42.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/3"}}},"commit":{"parents":[{"commit":"7a4cdd787f4e6f76d2c9cc32e82af433dc7e601d","subject":"Add a lock() context manager to image backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a4cdd787f4e6f76d2c9cc32e82af433dc7e601d"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-12 15:02:26.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-20 00:09:03.000000000","tz":0},"subject":"Rename Raw backend to Unbacked","message":"Rename Raw backend to Unbacked\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027Unbacked\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f8a60096202973edbc27a2c8bcc191d65e60a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f8a60096202973edbc27a2c8bcc191d65e60a37"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7a4cdd787f4e6f76d2c9cc32e82af433dc7e601d","is_merged_in_target_branch":false,"change_id":"Id9d469ac56ad61ff8273179c0edbf8a31f716e89","change_number":279625,"patch_set_number":3,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"35a32787b76c42bf91ef9bfac45c9422a2944440":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-02-23 17:12:08.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/4"}}},"commit":{"parents":[{"commit":"3b0536d283a41909373571438b2dd6a430c85f01","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b0536d283a41909373571438b2dd6a430c85f01"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-12 15:02:26.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-02-23 16:36:04.000000000","tz":0},"subject":"Rename Raw backend to Unbacked","message":"Rename Raw backend to Unbacked\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027Unbacked\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35a32787b76c42bf91ef9bfac45c9422a2944440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35a32787b76c42bf91ef9bfac45c9422a2944440"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3b0536d283a41909373571438b2dd6a430c85f01","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"920b00910eb4b585c42c5e6ae26c5093de6b9cce":{"kind":"REWORK","_number":5,"created":"2016-04-07 13:02:55.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/5"}}},"commit":{"parents":[{"commit":"7964e77b8057363f34d5cefec7730c1a011169c2","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7964e77b8057363f34d5cefec7730c1a011169c2"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:58:37.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/920b00910eb4b585c42c5e6ae26c5093de6b9cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/920b00910eb4b585c42c5e6ae26c5093de6b9cce"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7964e77b8057363f34d5cefec7730c1a011169c2","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d93a69ad58ddad492f19bcc400b8f055b81ba149":{"kind":"REWORK","_number":6,"created":"2016-04-12 14:56:03.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/6"}}},"commit":{"parents":[{"commit":"c55aae3b72745daa07d197c607c52f6b63146cc0","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c55aae3b72745daa07d197c607c52f6b63146cc0"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-12 14:09:36.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d93a69ad58ddad492f19bcc400b8f055b81ba149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d93a69ad58ddad492f19bcc400b8f055b81ba149"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c55aae3b72745daa07d197c607c52f6b63146cc0","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8bb8d51c308ad62e388b67fe64a08d421565ec19":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2016-04-29 12:33:30.000000000","uploader":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"ref":"refs/changes/26/279626/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/7"}}},"commit":{"parents":[{"commit":"6b701440a5bf78cd0116b9c22049e30a4f957b13","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b701440a5bf78cd0116b9c22049e30a4f957b13"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"paul-carlton2","email":"paul.carlton2@hpe.com","date":"2016-04-28 17:00:21.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bb8d51c308ad62e388b67fe64a08d421565ec19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bb8d51c308ad62e388b67fe64a08d421565ec19"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6b701440a5bf78cd0116b9c22049e30a4f957b13","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3e64d2985a6a532eb2d0888ba3defbfab7726951":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2016-05-03 13:15:31.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/8"}}},"commit":{"parents":[{"commit":"9b0e0f606f267a4f78a2d39b1355c2b26f33431e","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b0e0f606f267a4f78a2d39b1355c2b26f33431e"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-05-03 12:38:38.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e64d2985a6a532eb2d0888ba3defbfab7726951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e64d2985a6a532eb2d0888ba3defbfab7726951"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b0e0f606f267a4f78a2d39b1355c2b26f33431e","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7379f31f8d00114aded9b4dd54c87a915cc97181":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2016-05-05 14:24:07.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/9"}}},"commit":{"parents":[{"commit":"87b57384ae78fbd3327b1e777538906dd2b29f5c","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87b57384ae78fbd3327b1e777538906dd2b29f5c"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-05-04 13:53:42.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7379f31f8d00114aded9b4dd54c87a915cc97181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7379f31f8d00114aded9b4dd54c87a915cc97181"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"87b57384ae78fbd3327b1e777538906dd2b29f5c","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d957a2abade18646572ad462c76288389a0bcc83":{"kind":"REWORK","_number":10,"created":"2016-05-06 12:21:55.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/10"}}},"commit":{"parents":[{"commit":"09262b124dfb9cc4f037bcfb4ab78d6be85e2221","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09262b124dfb9cc4f037bcfb4ab78d6be85e2221"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-05-06 12:13:49.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d957a2abade18646572ad462c76288389a0bcc83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d957a2abade18646572ad462c76288389a0bcc83"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"09262b124dfb9cc4f037bcfb4ab78d6be85e2221","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"92ae93ce52fcf8f898d3f37a79d38a664f3a1eb5":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2016-05-09 14:37:22.000000000","uploader":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"ref":"refs/changes/26/279626/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/11"}}},"commit":{"parents":[{"commit":"4bfc8a6604158fff43b9c11acd6629967823cceb","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bfc8a6604158fff43b9c11acd6629967823cceb"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"paul-carlton2","email":"paul.carlton2@hpe.com","date":"2016-05-09 14:34:59.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/92ae93ce52fcf8f898d3f37a79d38a664f3a1eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/92ae93ce52fcf8f898d3f37a79d38a664f3a1eb5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4bfc8a6604158fff43b9c11acd6629967823cceb","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"00ded2de9a7d599ddb33fc5e11264b14d4506e7f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2016-05-12 07:52:28.000000000","uploader":{"_account_id":6608,"name":"Paul Carlton","email":"paul.carlton2@hpe.com","username":"paul-carlton2"},"ref":"refs/changes/26/279626/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/12"}}},"commit":{"parents":[{"commit":"b51d5d0c1f6f296bf2c4d3ce08cfbe010ba9f4e3","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b51d5d0c1f6f296bf2c4d3ce08cfbe010ba9f4e3"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"paul-carlton2","email":"paul.carlton2@hpe.com","date":"2016-05-12 07:49:08.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00ded2de9a7d599ddb33fc5e11264b14d4506e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00ded2de9a7d599ddb33fc5e11264b14d4506e7f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b51d5d0c1f6f296bf2c4d3ce08cfbe010ba9f4e3","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0ae1c33847a08f817783105b767af92a00ffd9f1":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2016-05-12 13:56:04.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/13"}}},"commit":{"parents":[{"commit":"e793162e078acbdbe7c588a9f1392111ece38987","subject":"Rename Image.check_image_exists to Image.exists()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e793162e078acbdbe7c588a9f1392111ece38987"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-05-12 13:55:12.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nImplements: blueprint libvirt-instance-storage\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ae1c33847a08f817783105b767af92a00ffd9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ae1c33847a08f817783105b767af92a00ffd9f1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e793162e078acbdbe7c588a9f1392111ece38987","is_merged_in_target_branch":false,"change_id":"Ic15e8412971b2a3f8e87ec68605c4753e23f94b5","change_number":270998,"patch_set_number":18,"change_status":"MERGED"}],"branch":"refs/heads/master"},"03289d3cf2b46428619299b73bd5acb09228a486":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2016-05-13 14:28:52.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/14"}}},"commit":{"parents":[{"commit":"93a475cd54ee2286ff21701fa06c5f0b6395ad52","subject":"Remove fake_imagebackend.Raw and cleanup dependent tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93a475cd54ee2286ff21701fa06c5f0b6395ad52"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-05-13 14:27:21.000000000","tz":60},"subject":"Rename Raw backend to NoBacking","message":"Rename Raw backend to NoBacking\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027NoBacking\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027nobacking\u0027. The effect of this is identical to setting the value\n\u0027raw\u0027.\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03289d3cf2b46428619299b73bd5acb09228a486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03289d3cf2b46428619299b73bd5acb09228a486"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"93a475cd54ee2286ff21701fa06c5f0b6395ad52","is_merged_in_target_branch":true,"change_id":"Iffdc0c66f1ec4428b98fffc807af636dd491a080","change_number":267661,"patch_set_number":20,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cff4d78a4a7ddbb18cb875c66b3c56f04a6caea3":{"kind":"REWORK","_number":15,"created":"2016-05-17 11:03:59.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/26/279626/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/279626/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/279626/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/279626/15"}}},"commit":{"parents":[{"commit":"52a831e78e3bfbbaa1e129ed931a2d2b0dc48071","subject":"Merge \"Switch api unit tests to use v2.1 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/52a831e78e3bfbbaa1e129ed931a2d2b0dc48071"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-04-07 12:40:58.000000000","tz":60},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2016-05-17 09:51:38.000000000","tz":60},"subject":"Rename Raw backend to Flat","message":"Rename Raw backend to Flat\n\nAs mentioned in a comment (which this patch deletes), calling this\nclass \u0027Raw\u0027 was confusing, because it is not always raw. It is also a\nsource of bugs, because some code assumes that the format is always\nraw, which it is not. This patch does not fix those bugs.\n\nWe rename it to \u0027Flat\u0027, which describes it accurately. We also add\ndoctext describing what it does.\n\nDocImpact\n\nThe config option libvirt.images_type gets an additional value:\n\u0027flat\u0027. The effect of this is identical to setting the value \u0027raw\u0027.\n\nChange-Id: I93f0a2cc568b60c2b3f7509449167f03c3f30fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cff4d78a4a7ddbb18cb875c66b3c56f04a6caea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cff4d78a4a7ddbb18cb875c66b3c56f04a6caea3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a831e78e3bfbbaa1e129ed931a2d2b0dc48071","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
