)]}'
{"id":"openstack%2Fnova~305369","triplet_id":"openstack%2Fnova~master~I2c081878c18f2905a9509cfdbc84deec32127154","project":"openstack/nova","branch":"master","topic":"versioned_response","hashtags":[],"change_id":"I2c081878c18f2905a9509cfdbc84deec32127154","subject":"Pin HTTP response codes to API microversions","status":"ABANDONED","created":"2016-04-13 15:57:37.000000000","updated":"2016-11-23 22:53:28.000000000","total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b546c2e4dfc05722fa640ed29a953e367659c4d","_number":305369,"virtual_id_number":305369,"owner":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2016-09-08 11:10:34.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"date":"2016-09-08 11:49:23.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-09-21 02:52:27.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-09-08 18:09:22.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-09-08 10:56:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2016-09-08 10:17:18.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"date":"2016-09-09 08:21:35.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-09-08 12:12:25.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-09-08 12:00:51.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":1,"date":"2016-09-08 12:40:15.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-10-04 23:43:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2016-09-08 09:55:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-04-19 14:35:37.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2016-05-20 18:49:37.000000000","updated_by":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"reviewer":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"state":"REVIEWER"},{"updated":"2016-06-01 13:05:34.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2016-06-09 13:48:15.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2016-06-09 13:49:05.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2016-06-22 17:39:34.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2016-06-24 15:12:53.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2016-06-29 13:25:48.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-29 13:31:34.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2016-07-04 18:50:13.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-07-06 13:21:32.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-09-06 16:45:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 10:17:18.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 10:56:13.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 11:10:34.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 11:49:23.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 12:00:51.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 12:12:25.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 12:40:15.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-08 18:09:22.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-09 08:21:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-21 02:52:27.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-10-04 23:43:31.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"}],"messages":[{"id":"8b46d4764d67ccf2e1fd2cc41ddfcd5e8bc93cb1","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-04-13 15:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cdb3b59f494d79505a881f431a3de776b93eb5d","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-13 16:15:21.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/902efd7cdbd84e199cf448c15f4e91f7/index.html : SUCCESS in 16m 20s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b0b716183495431fa24be4d48eb0fd0e/index.html : SUCCESS in 15m 06s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c357b769aa3f4c52b81391345bf70273/index.html : SUCCESS in 15m 08s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/1/check/real-db-upgrade_nova_percona_user_001:th-percona/9860955b1c7f4df0814c157330b2cb8c/index.html : SUCCESS in 15m 41s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"1cad307ef0c183ae6cafe54f0d1e911019052328","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-13 16:18:42.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/305369/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"62789e9651ae7fb242fb7edff0aebd024f3c3645","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-04-13 17:05:50.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/69/305369/1/check/check-ibm-tempest-dsvm-full/736f7f3/ : SUCCESS in 1h 07m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da51db139d1147a4321324e9cada572b467d0004","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-13 17:06:54.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/1/12132/index.html : SUCCESS in 01 h 07 m 13 s","accounts_in_message":[],"_revision_number":1},{"id":"d6d25f267d577ed2f2e6f8cce19c91c329c7fa49","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-13 17:08:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/1/ : FAILURE in 1h 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f2fa60b5b6751520d77eebf4f40f982600d6dcf7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-13 17:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/69/305369/1/check/gate-nova-docs/dd7da20//doc/build/html/ : SUCCESS in 2m 38s\n- gate-nova-pep8 http://logs.openstack.org/69/305369/1/check/gate-nova-pep8/39c6bbd/ : SUCCESS in 4m 29s\n- gate-nova-python27-db http://logs.openstack.org/69/305369/1/check/gate-nova-python27-db/ffff119/ : SUCCESS in 10m 31s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/1/check/gate-nova-python34-db/380dbd9/ : SUCCESS in 10m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-full/7bb63da/ : SUCCESS in 30m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-postgres-full/c7a2adb/ : SUCCESS in 33m 18s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-neutron-full/6aa5155/ : SUCCESS in 53m 16s\n- gate-grenade-dsvm http://logs.openstack.org/69/305369/1/check/gate-grenade-dsvm/9977062/ : SUCCESS in 36m 31s\n- gate-nova-tox-db-functional http://logs.openstack.org/69/305369/1/check/gate-nova-tox-db-functional/a2ba86f/ : SUCCESS in 10m 21s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/1/check/gate-grenade-dsvm-multinode/9825b69/ : SUCCESS in 47m 36s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/91afd7e/ : SUCCESS in 36m 29s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-cells/6c13267/ : SUCCESS in 26m 36s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-nv http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-nv/5f3686f/ : SUCCESS in 34m 07s (non-voting)\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-multinode-full/1cbaa40/ : SUCCESS in 43m 03s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/69/305369/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/4c208cb/ : SUCCESS in 1h 06m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad1cea8767dc1304237046c775387f4f0d238af1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-13 17:48:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/1/check/dsvm-tempest-xen/34271d6 : FAILURE in 1h 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"93985740c9f4564bf69f71a05d380c1f0ea6a8f1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-13 18:25:54.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-69-305369-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"145dbe23a633002a3851d1ccccce6248073a4386","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-04-13 23:09:30.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/69/305369/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"f6bb33592786885816913ebac6325bc9d20b0cc7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-14 00:23:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/1/check/check-dsvm-tempest-vz7-exe-minimal/0ed239e : SUCCESS in 1h 24m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"a066de3538959940c35dabc972d8617f4729df6c","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-04-14 08:29:11.000000000","message":"Patch Set 1:\n\nxen: recheck","accounts_in_message":[],"_revision_number":1},{"id":"66be6f79a6f57aebb035d0c10667fc64601e292a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-14 09:47:37.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/1/check/dsvm-tempest-xen/7771ee3 : FAILURE in 1h 17m 25s","accounts_in_message":[],"_revision_number":1},{"id":"464af6a70fa2159a38177667155dabdfb374ddc9","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-04-14 14:18:32.000000000","message":"Patch Set 1:\n\nxen: recheck","accounts_in_message":[],"_revision_number":1},{"id":"34b3df47ebf0f7e25a3040de25c8448d4fdae771","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-14 16:43:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/1/check/dsvm-tempest-xen/2118f10 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"beaabff89c86b6aabf92f3c5cc7ddf100d31bbd1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-15 02:06:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/305369/1/3941 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6b19e7342a1ad6aa280fa429462a4d290e400f27","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2016-04-15 20:25:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8378dbff7c6bef7cbdc713e49837f410edf34e6a","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-04-19 14:11:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"915e9429ed497ecc525f76f2c03760690cb75629","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-04-19 14:33:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29e36687c14091be38b5bc02e93a2cbd5cedb9f4","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2016-04-19 14:35:37.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good!","accounts_in_message":[],"_revision_number":2},{"id":"8b14fc479308d5dab8d807a18274e4f61f037377","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-04-19 14:46:29.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/0d2a793e14f44ee2a4c362eba30e27ad/index.html : SUCCESS in 11m 30s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6bb9ceb1364c4302b29be4e8c806c019/index.html : SUCCESS in 10m 00s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/a092a8f810d44560aa3a1bf348c9514b/index.html : SUCCESS in 10m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/2/check/real-db-upgrade_nova_percona_user_001:th-percona/fc9be8eaf1da4273a5b53ec741e6a63e/index.html : SUCCESS in 9m 17s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"a8e1d07c5b5185e15f8bd638f18a7b7610bb5016","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-04-19 15:41:38.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/305369/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"89b50d5cf47fbb6ab2360b11752384b8ea08c7bc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-04-19 15:49:20.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/2/12668/index.html : SUCCESS in 01 h 12 m 19 s","accounts_in_message":[],"_revision_number":2},{"id":"08562b665bd7aa8667ceb9d17819cc0845faaacf","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-04-19 15:54:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/2/check/dsvm-tempest-xen/34c11fa : SUCCESS in 1h 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2097d7d9eebc02808481eb681f2e5ad04d9f5cb7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-04-19 16:17:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/2/ : FAILURE in 1h 44m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f180c78b1738a06b2850eddd9f0cc917e9131b69","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-04-19 23:24:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/2/check/check-dsvm-tempest-vz7-exe-minimal/b090feb : SUCCESS in 1h 14m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"d064516fefd0fa90cdcec1686fe0324bd2ba668c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-20 00:21:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/69/305369/2/check/gate-nova-docs/55d3145//doc/build/html/ : SUCCESS in 4m 58s\n- gate-nova-pep8 http://logs.openstack.org/69/305369/2/check/gate-nova-pep8/08c1db8/ : SUCCESS in 3m 15s\n- gate-nova-python27-db http://logs.openstack.org/69/305369/2/check/gate-nova-python27-db/b08f5e8/ : SUCCESS in 11m 26s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/2/check/gate-nova-python34-db/e02cb74/ : SUCCESS in 12m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-full/09cd4b7/ : SUCCESS in 34m 25s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-postgres-full/d0b06dc/ : SUCCESS in 40m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-neutron-full/95bf425/ : SUCCESS in 1h 13m 18s\n- gate-grenade-dsvm http://logs.openstack.org/69/305369/2/check/gate-grenade-dsvm/edf2f4f/ : SUCCESS in 37m 27s\n- gate-nova-tox-db-functional http://logs.openstack.org/69/305369/2/check/gate-nova-tox-db-functional/879d96c/ : SUCCESS in 13m 26s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/2/check/gate-grenade-dsvm-multinode/88abe2b/ : SUCCESS in 27m 54s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/9a8c8bf/ : FAILURE in 44m 23s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-cells/5c7fcf5/ : SUCCESS in 34m 56s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/50eb31d/ : SUCCESS in 35m 52s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-multinode-full/8220b45/ : SUCCESS in 45m 08s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/69/305369/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/1fb8d18/ : SUCCESS in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a39c50266b8df9250ca830fd41254dd1e4145238","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-04-20 02:19:02.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-69-305369-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"5b255d093349acdf889bac7940c97996877f2529","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-04-20 07:00:08.000000000","message":"Patch Set 2:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/69/305369/2 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":2},{"id":"bd1ab6c7504151b4cd94c3a3a441b632c0eb7f3d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-04-20 18:00:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/305369/2/4138 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"14158a0d2184f7acd9ac9aa16f9ca4a033c4f197","author":{"_account_id":6802,"name":"Joel Coffman","email":"jmc7tp@gmail.com","username":"joel-coffman"},"date":"2016-05-20 18:49:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7e068ec2b331aa51e50e45f959cf6b8217728469","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2016-06-01 13:05:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e268841035e0f90ffe6e155c1ba99b007164f9c3","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-07 16:20:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6c45c50f40cc27248bc860e77cde34d09219f25","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-07 16:43:13.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/65017141bdc44f4bb2f5c631b59ed702/index.html : SUCCESS in 17m 08s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/fe81a8b68e2e4a11ae34af8059bd8fcc/index.html : SUCCESS in 20m 59s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/acc9d98497f94a2daec26b2008361e43/index.html : SUCCESS in 18m 43s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/3/check/real-db-upgrade_nova_percona_user_001:th-percona/39ab1ec902da4ebabef1c5354fb71418/index.html : SUCCESS in 19m 42s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"9c56e42a67fae41b38e790474056b9085dab8174","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-07 17:34:52.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/3/17936/index.html : SUCCESS in 01 h 12 m 43 s","accounts_in_message":[],"_revision_number":3},{"id":"1cecee4a4831d4c6bd98c9c947686cbb70a6d23d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-07 17:37:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/3/check/dsvm-tempest-xen/3210417 : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1420028ca560c1b77d955d8fd890b6a82719b25d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-07 18:35:01.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/69/305369/3/check/check-ibm-tempest-dsvm-full/70e682b/ : SUCCESS in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"759e9a6791b8744f2f173496c465fc09d9af10b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 18:57:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/69/305369/3/check/gate-nova-docs/ba7eb17//doc/build/html/ : SUCCESS in 3m 59s\n- gate-nova-pep8 http://logs.openstack.org/69/305369/3/check/gate-nova-pep8/eebe2bc/ : SUCCESS in 6m 09s\n- gate-nova-python27-db http://logs.openstack.org/69/305369/3/check/gate-nova-python27-db/ece4b81/ : SUCCESS in 8m 39s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/3/check/gate-nova-python34-db/fa92795/ : SUCCESS in 10m 45s\n- gate-tempest-dsvm-full http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-full/efeb649/ : SUCCESS in 47m 44s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-postgres-full/3c460e2/ : SUCCESS in 50m 29s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-neutron-full/5bcd68f/ : SUCCESS in 1h 18m 39s\n- gate-grenade-dsvm http://logs.openstack.org/69/305369/3/check/gate-grenade-dsvm/83c53ef/ : SUCCESS in 47m 05s\n- gate-nova-tox-db-functional http://logs.openstack.org/69/305369/3/check/gate-nova-tox-db-functional/0e0991e/ : SUCCESS in 11m 16s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/3/check/gate-grenade-dsvm-multinode/3512f76/ : SUCCESS in 1h 14m 33s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8ddd3c8/ : SUCCESS in 39m 15s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-cells/496bfe6/ : SUCCESS in 43m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/a519e1c/ : SUCCESS in 49m 47s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-multinode-full/db6f0df/ : FAILURE in 1h 05m 32s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/69/305369/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/fb48f11/ : SUCCESS in 54m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f659d0474ffdb2e5a0d052c43c296a01d4eef3b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-07 20:26:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/3/ : SUCCESS in 2h 56m 00s","accounts_in_message":[],"_revision_number":3},{"id":"2a5e95e1ffedaf9679cc28654a3a7c3b60528b32","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-07 21:22:54.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/305369/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"373474f04b6bd78e4e87652291201762f82d3413","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-07 21:52:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/69/305369/3/check/tempest-dsvm-full-nfv/0793687 : SUCCESS in 1h 26m 38s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/69/305369/3/check/tempest-dsvm-intel-nfv/3e91006 : FAILURE in 36m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1cf44d24084d3bfecbd3471aa3a3d67d1d30c849","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-08 02:11:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/69/305369/3/check/dsvm-tempest-kvm/accd757 : SUCCESS in 1h 08m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"eff20fd00de424cd912c7acbddb0d6f81a591c82","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-08 07:33:10.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/3/check/check-dsvm-tempest-vz7-exe-minimal/a8dd8b1 : FAILURE in 2h 28m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"7a72f4c5964ce1b75c9dfd96c453a8b680569a2c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-09 00:48:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/69/305369/3/check-nova/Nova-ML2-Sriov/075bd14 : SUCCESS in 32m 51s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/69/305369/3/check-nova/Nova-MACVTAP-ML2-Sriov/598ba2d : SUCCESS in 34m 24s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"8710bbcb00f863f79adc6aaac2f990ebd55d26d9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-06-09 05:03:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n[http://file01.eng.vmware.com/ci-logs/dsvm-ext-nova/305369/3/230/ dsvm-ext-nova] : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"8530392bcdae1b83530842b75f01e09f20eebe23","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-06-22 16:01:50.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cea9552c8232fc356f7ccb482d999de4a84efb65","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2016-06-22 17:39:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nA few minor suggestions, but overall this looks excellent, thank you Gleb!\n\n-jay","accounts_in_message":[],"_revision_number":3},{"id":"50edd9ce8340f556e89ae2a762032a0b54a57579","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-06-23 12:05:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1a078f1990f502b0a765d933baacfb6340e7268d","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-23 15:08:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a40bfc38bd61a7fa9a66ed8a79970fe5286fd14c","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-06-23 15:14:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38e4d0434336eb2d81aaf5cb3f84fc88d64b505f","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-24 11:25:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b7a8cffb1fe7c6b38549ebf79c95c8411384720","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-24 11:55:04.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f41d8a65ed374b63a36499bc0d75f85d/index.html : SUCCESS in 16m 17s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/23276a3287a5424d90d4c07aadb0c38f/index.html : SUCCESS in 17m 34s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/9a0cdf45f1d74445802ab0bda8e231c1/index.html : SUCCESS in 27m 39s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/4/check/real-db-upgrade_nova_percona_user_001:th-percona/747bb012b719425cb3d0b1bc7ad9de02/index.html : SUCCESS in 22m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"f69456ea1c5cc23f25ead9a09f4a68b88cb2dc5a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-24 12:45:02.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/4/21401/index.html : SUCCESS in 01 h 15 m 38 s","accounts_in_message":[],"_revision_number":4},{"id":"f4ac9bc5e3e4ba58365d7cfcf6054c1dcc00d384","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-24 13:39:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/4/ : SUCCESS in 1h 52m 09s","accounts_in_message":[],"_revision_number":4},{"id":"ff8b7c3e06c5ceb0bb2ec713ced92c3921162299","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-24 14:02:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/4/check/dsvm-tempest-xen/113e60f : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":4},{"id":"b3a10637095302948a3c717e551f306ace6b9f4f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-24 14:07:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/69/305369/4/check/gate-nova-docs/3cdc8ca//doc/build/html/ : SUCCESS in 3m 01s\n- gate-nova-pep8 http://logs.openstack.org/69/305369/4/check/gate-nova-pep8/6f4f3a7/ : FAILURE in 6m 01s\n- gate-nova-python27-db http://logs.openstack.org/69/305369/4/check/gate-nova-python27-db/a92e4c1/ : SUCCESS in 10m 07s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/4/check/gate-nova-python34-db/7679a1f/ : SUCCESS in 8m 26s\n- gate-tempest-dsvm-full http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-full/a0fd669/ : SUCCESS in 48m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-postgres-full/716747b/ : SUCCESS in 51m 52s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-neutron-full/9310f96/ : SUCCESS in 56m 34s\n- gate-grenade-dsvm http://logs.openstack.org/69/305369/4/check/gate-grenade-dsvm/b9720f8/ : SUCCESS in 45m 27s\n- gate-nova-tox-db-functional http://logs.openstack.org/69/305369/4/check/gate-nova-tox-db-functional/6be58d0/ : SUCCESS in 14m 18s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/4/check/gate-grenade-dsvm-multinode/7c7d70f/ : FAILURE in 2h 41m 29s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b2fb459/ : SUCCESS in 35m 29s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-cells/5734237/ : SUCCESS in 42m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph/7465122/ : SUCCESS in 47m 15s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-multinode-full/980a989/ : SUCCESS in 52m 00s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/e9a22f1/ : SUCCESS in 1h 01m 17s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/69/305369/4/check/gate-tempest-dsvm-multinode-live-migration/3c46de6/ : SUCCESS in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9fc0de473044fd8867ed70a690d6db2c7637ea23","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-06-24 15:12:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dfb60372c3b8f89de4673c8d402324f2a7f0b07a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-24 15:48:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/69/305369/4/check/tempest-dsvm-full-nfv/618b6d2 : SUCCESS in 1h 53m 16s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/69/305369/4/check/tempest-dsvm-intel-nfv/4aa008b : SUCCESS in 35m 42s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/69/305369/4/check/tempest-dsvm-ovsdpdk-nfv-networking/5e21a96 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":4},{"id":"c991cc1c49547393542562fef7f049d556cbdcfd","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-25 06:46:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/69/305369/4/check/dsvm-tempest-kvm/02f6784 : FAILURE in 1h 18m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"0ad9fddee3c9cdd1226a44f17cfad65031ab7551","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-25 07:01:34.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/305369/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"540183c35414c321189e8aa95c3be6fde99bee15","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-25 07:41:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/4/check/check-dsvm-tempest-vz7-exe-minimal/a69b715 : SUCCESS in 45m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"cd96776cdd131ca857bf7d4bbc889b91f61fc970","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-27 00:21:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/69/305369/4/check-nova/Nova-ML2-Sriov/a2514f4 : SUCCESS in 31m 32s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/69/305369/4/check-nova/Nova-MACVTAP-ML2-Sriov/9e12b12 : SUCCESS in 32m 09s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"e50e5eb8272edf6d1f81a7c917be3cff817f244b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-06-27 07:15:52.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n[http://file01.eng.vmware.com/ci-logs/dsvm-ext-nova/305369/4/781/ dsvm-ext-nova] : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7524af416537782e4559d47fc0588e758e29c53e","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-27 13:13:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"010ad705cd2a2dc47cc1912cfedda67322ae21a8","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-27 13:45:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c564b3e4b1c43f32d93a4357d1684062da9d70b1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-27 14:44:42.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/555def53a9bc431e8b417124a89dbc12/index.html : SUCCESS in 19m 52s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b93e1d80f1c1465dab4c7e4352b570cb/index.html : SUCCESS in 23m 14s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7e5dcf7bae4a49f784379eeef625e6fc/index.html : SUCCESS in 16m 38s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/5/check/real-db-upgrade_nova_percona_user_001:th-percona/e2e1daedf951447f9f2a664e3d1b0cca/index.html : SUCCESS in 25m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"a7655029cfb607de66104ac14977b94252f2d899","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-27 15:09:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/69/305369/5/check/gate-nova-docs/7f41fd1//doc/build/html/ : SUCCESS in 5m 28s\n- gate-nova-pep8 http://logs.openstack.org/69/305369/5/check/gate-nova-pep8/e0d9401/ : SUCCESS in 4m 55s\n- gate-nova-python27-db http://logs.openstack.org/69/305369/5/check/gate-nova-python27-db/639bd70/ : FAILURE in 10m 46s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/5/check/gate-nova-python34-db/acb7f85/ : FAILURE in 12m 13s\n- gate-tempest-dsvm-full http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-full/8edab0e/ : SUCCESS in 47m 11s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-postgres-full/d793f45/ : SUCCESS in 45m 46s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-neutron-full/3cf7ad0/ : SUCCESS in 1h 00m 08s\n- gate-grenade-dsvm http://logs.openstack.org/69/305369/5/check/gate-grenade-dsvm/dbce638/ : SUCCESS in 43m 28s\n- gate-nova-tox-db-functional http://logs.openstack.org/69/305369/5/check/gate-nova-tox-db-functional/edb2f94/ : SUCCESS in 16m 27s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/5/check/gate-grenade-dsvm-multinode/6efe883/ : SUCCESS in 46m 34s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/4dea749/ : SUCCESS in 22m 01s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-cells/aa02545/ : SUCCESS in 29m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph/65fee39/ : SUCCESS in 39m 14s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-multinode-full/659db5f/ : SUCCESS in 1h 02m 30s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/ee151bc/ : SUCCESS in 48m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/69/305369/5/check/gate-tempest-dsvm-multinode-live-migration/c98f0a6/ : SUCCESS in 43m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"553afb4e4dd66395f3bd8b31585e8a1c4e1aab30","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-27 15:35:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/5/ : SUCCESS in 1h 35m 41s","accounts_in_message":[],"_revision_number":5},{"id":"33ec3aebc63a1fa157ae12493aea1ab474941044","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-27 15:40:10.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/5/21771/index.html : SUCCESS in 01 h 10 m 55 s","accounts_in_message":[],"_revision_number":5},{"id":"ad494f53adff9a93a47f7b0208e7770bbffa0922","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-27 16:14:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/5/check/dsvm-tempest-xen/cf8a79b : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":5},{"id":"1bcd5d26d36002196e2112d26488af9a50bf652d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-27 18:03:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/69/305369/5/check/tempest-dsvm-full-nfv/787fc20 : SUCCESS in 1h 24m 54s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/69/305369/5/check/tempest-dsvm-intel-nfv/54da686 : SUCCESS in 40m 18s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/69/305369/5/check/tempest-dsvm-ovsdpdk-nfv-networking/5a9919d : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":5},{"id":"c8e6dfcdb0cfe396d797e0d165b65226d0b40ab6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-27 18:47:06.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/305369/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e135dcbfccd0d8186c75216f8b38b50e76aadb3e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-27 20:41:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/5/check/check-dsvm-tempest-vz7-exe-minimal/25e5b99 : SUCCESS in 1h 56m 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"01e4d28611f4d498b4f3ed6308947f80b7c091d5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-27 21:39:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/69/305369/5/check/dsvm-tempest-kvm/f712e3c : SUCCESS in 1h 21m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"7ecf1db29601639bc04030f4286c0282a40fae4d","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-29 12:57:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6d9dc6a822dbca89cf443c4593d4280f115a8a7c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-29 13:25:48.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/83dd607443294744986ed4e2d03a447c/index.html : SUCCESS in 18m 32s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/aebbf60e165f4b1a9df58e9bd182b3aa/index.html : SUCCESS in 15m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/69/305369/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/17be50c0375c457e8ca79309fa23c4f0/index.html : SUCCESS in 24m 37s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/69/305369/6/check/real-db-upgrade_nova_percona_user_001:th-percona/89272ba18e4f486fafa158cdbe406673/index.html : SUCCESS in 20m 10s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"078b8390078c159b640009c70119135fa4276436","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-29 13:29:05.000000000","message":"Patch Set 6:\n\nI\u0027m not sure I\u0027m following why we need this - we\u0027ve been getting around this by adding versioned methods which can have their own response code and own expected_errors decorator, e.g.:\n\nhttps://review.openstack.org/#/c/326940/16/nova/api/openstack/compute/hypervisors.py","accounts_in_message":[],"_revision_number":6},{"id":"81bdd3aadc8f96bd54f70a484cb2b16af47e4813","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-29 13:31:34.000000000","message":"Patch Set 6: Code-Review-1\n\nI\u0027d rather we be explicit about the microversions on the methods themselves, along with the expected errors and response codes. We also require a microversion in most cases when a response code changes or is added to a method:\n\nhttp://docs.openstack.org/developer/nova/api_microversion_dev.html#when-do-i-need-a-new-microversion\n\nI\u0027d rather we were explicit in the methods and fix bugs if we miss something in review rather than add a decorator that does this implicitly for us - which I\u0027ll forget is there but things magically happen.","accounts_in_message":[],"_revision_number":6},{"id":"e43858fe4857a693c8b81f747640ff9669eb091e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-29 13:54:51.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/6/ : FAILURE in 51m 22s","accounts_in_message":[],"_revision_number":6},{"id":"5a39fcdb2be10e1bb13f5bdcccd609641b605f73","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-29 13:57:42.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/69/305369/6/check/check-ibm-tempest-dsvm-full/275959d/ : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"32c835d6c4dbaca3e57899b553ab4029e5fdaf89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-29 14:11:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/69/305369/6/check/gate-nova-docs/d16db8c//doc/build/html/ : SUCCESS in 4m 52s\n- gate-nova-pep8 http://logs.openstack.org/69/305369/6/check/gate-nova-pep8/50ab221/ : SUCCESS in 6m 19s\n- gate-nova-python27-db http://logs.openstack.org/69/305369/6/check/gate-nova-python27-db/0b02de9/ : SUCCESS in 14m 19s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/6/check/gate-nova-python34-db/5d22e87/ : SUCCESS in 10m 30s\n- gate-tempest-dsvm-full http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-full/13ad23c/ : SUCCESS in 48m 02s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-postgres-full/44e7f19/ : SUCCESS in 49m 58s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-neutron-full/d5bf29b/ : SUCCESS in 1h 03m 14s\n- gate-grenade-dsvm http://logs.openstack.org/69/305369/6/check/gate-grenade-dsvm/0a466f0/ : SUCCESS in 25m 06s\n- gate-nova-tox-db-functional http://logs.openstack.org/69/305369/6/check/gate-nova-tox-db-functional/576e778/ : SUCCESS in 16m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/6/check/gate-grenade-dsvm-multinode/1722a06/ : SUCCESS in 50m 07s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b6a0a35/ : SUCCESS in 21m 33s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-cells/b91f82d/ : SUCCESS in 39m 27s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph/0cd9e04/ : FAILURE in 45m 11s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-multinode-full/7dc90b5/ : SUCCESS in 54m 41s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/8a5f6bc/ : SUCCESS in 51m 53s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-multinode-live-migration/fba498f/ : FAILURE in 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e05b6248e90351921bea75cd40d58ecb1c1593cf","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-29 14:28:54.000000000","message":"Patch Set 6: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/6/22318/index.html : SUCCESS in 01 h 06 m 57 s","accounts_in_message":[],"_revision_number":6},{"id":"9e691873975b2258e01e213157ac71e5cba4e065","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-29 14:37:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/6/check/dsvm-tempest-xen/03d0dd5 : FAILURE in 1h 23m 14s","accounts_in_message":[],"_revision_number":6},{"id":"3d961cba278dfa18e2f4caf8b948daa62aa49fa3","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-29 14:43:56.000000000","message":"Patch Set 6:\n\nMatt Riedemann, but in that case we still have only one possible version of response code and there is no backward compatibility for older versions of nova clients. But this decorator help us to handle this gracefully.","accounts_in_message":[],"_revision_number":6},{"id":"bd9d43ee1ad6eeaeb645604eec4c00ceedb9e3a6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-29 14:48:06.000000000","message":"Patch Set 6:\n\nGleb, I don\u0027t worry about changing response codes in a microversion, regardless of backward compatibility. That\u0027s why we have microversions, and that\u0027s why the client opts into them - they should have a clear understanding of the changes in the microversion.","accounts_in_message":[],"_revision_number":6},{"id":"f7e92b5af73f743704dde6595ccb6cbe1f2e608b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-29 15:04:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/69/305369/6/check-nova/Nova-ML2-Sriov/a335638 : SUCCESS in 32m 26s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/69/305369/6/check-nova/Nova-MACVTAP-ML2-Sriov/8e97abc : NOT_REGISTERED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"896b3f39407792d7eb7030779e17494db7ee459e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-29 15:24:07.000000000","message":"Patch Set 6:\n\nBTW, if we have microversions that change the response code those should be clearly documented in the nova REST API history docs, it\u0027s a bug if they don\u0027t.","accounts_in_message":[],"_revision_number":6},{"id":"ab6d96ffb6b80d289ad330e0e2284961afa9628f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-29 18:48:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-full-nfv/9aa1654 : SUCCESS in 1h 42m 17s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-intel-nfv/6c14770 : SUCCESS in 38m 44s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-ovsdpdk-nfv-networking/dab77de : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":6},{"id":"b602ba693b7266ac9497a887c6b6832620781ec8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-29 19:40:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/69/305369/6/check/dsvm-tempest-kvm/aa1424e : SUCCESS in 56m 08s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"4ceb00f788c7a36a76016c0c696579a91a6d11ba","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-29 22:00:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/6/check/check-dsvm-tempest-vz7-exe-minimal/3ce02ca : SUCCESS in 1h 11m 08s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"95936afab2cb78e46304e98f77e922b78a53b402","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-06-30 17:26:48.000000000","message":"Patch Set 6:\n\nYeah, but i consider response code of api method as natural part of method like arguments, response object etc. All this stuff is versioned, so if you add new argument to method you should bump microversion and probably leave old method for older versions.  So response code is like return value in functions for HTTP (together with response if it exists). My point was, if we have verioning of method signatires, why dont we have versioning of http methods?","accounts_in_message":[],"_revision_number":6},{"id":"0e3e3eab07e412140174a3452930d4062ead7839","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-07-03 12:52:15.000000000","message":"Patch Set 6:\n\nI agree that we shouldn\u0027t change the response code outside of a microversion - did that happen in a specific case that\u0027s not fixed? When we need to change expected_errors or a response in a given microversion, we should be splitting out into multiple versioned methods explicitly.","accounts_in_message":[],"_revision_number":6},{"id":"69e7c87f8b952cc8e5587a1acfef8d60f77aaa32","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-04 18:50:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/69/305369/6/check/dsvm-tempest-xen/b6cd0ef : SUCCESS in 1h 17m 15s","accounts_in_message":[],"_revision_number":6},{"id":"721b6270780484678b1aad5d8343d020fc4e38c6","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-07-06 13:21:32.000000000","message":"Patch Set 6:\n\n+1 matt\u0027s questions.\n\nThe intent in the code is to version the full HTTP request and response pair, status code, method and all.\n\nI think style folks seems it best to keep with a separate method if the return code changes, so its nice and explicit.","accounts_in_message":[],"_revision_number":6},{"id":"db1c364ce1d242a892df619f92de719ed6baad47","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-07-12 13:31:36.000000000","message":"Patch Set 6:\n\nMy point why i prefer this solution over https://github.com/openstack/nova/blob/master/nova/api/openstack/wsgi.py#L915 is that we dont need to duplicate code each time when response code is changed. We can multiple decorators response_versioned on controller function and for each version we will get appropriate response code.","accounts_in_message":[],"_revision_number":6},{"id":"dcb4ffcc339d0a3f91de5c4f2d6d5b2a80bcf361","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-09-06 14:49:21.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d97aaeb08deda04adf1c9d3a5ca47ffc9abb2085","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-06 15:29:53.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/6/ : FAILURE in 40m 09s","accounts_in_message":[],"_revision_number":6},{"id":"677e64a60c68e3056da2a1af0a7c37ec14f509b5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-06 15:50:24.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/69/305369/6/check/check-ibm-tempest-dsvm-full/a860a93/ : SUCCESS in 1h 00m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"62fb48b6546529e8fe9ba3e7440ac0b7492241e1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-06 16:06:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/69/305369/6/check/gate-nova-docs-ubuntu-xenial/6e886e2//doc/build/html/ : SUCCESS in 3m 55s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-nova-pep8-ubuntu-xenial/f78a308/ : SUCCESS in 6m 38s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-nova-python27-db-ubuntu-xenial/b518e42/ : SUCCESS in 16m 04s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/6/check/gate-nova-python34-db/d85f2b9/ : SUCCESS in 13m 49s\n- gate-nova-python35-db-nv http://logs.openstack.org/69/305369/6/check/gate-nova-python35-db-nv/fcd2133/ : SUCCESS in 15m 57s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/69/305369/6/check/gate-grenade-dsvm-neutron-ubuntu-trusty/ded4101/ : SUCCESS in 46m 32s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/4e837ee/ : SUCCESS in 1h 05m 39s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-full-ubuntu-xenial/a0bdab6/ : SUCCESS in 50m 51s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/69/305369/6/check/gate-grenade-dsvm-ubuntu-trusty/087a5a2/ : SUCCESS in 39m 23s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-nova-tox-db-functional-ubuntu-xenial/cd99abd/ : SUCCESS in 15m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/6/check/gate-grenade-dsvm-multinode/b8cddcb/ : SUCCESS in 47m 30s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/9e43b81/ : SUCCESS in 34m 48s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-cells/6420768/ : SUCCESS in 36m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph/7f0fd48/ : SUCCESS in 41m 28s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/22d60ae/ : SUCCESS in 1h 15m 22s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/eaa2927/ : SUCCESS in 1h 08m 02s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/69/305369/6/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/b782848/ : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":6},{"id":"729fa346aad575896ad7907f70be2dbe63d931dd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-06 16:45:31.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-69-305369-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"7bd8e29ef2abaf3a520a93591208b9a6bdc44dac","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-06 19:06:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-full-nfv/9f5c2ce : SUCCESS in 2h 07m 04s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-intel-nfv/f8af825 : SUCCESS in 49m 57s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-ovsdpdk-nfv-networking/50a667a : SUCCESS in 1h 03m 10s\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-full-nfv-xenial/db36c09 : SUCCESS in 2h 12m 40s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-intel-nfv-xenial/6cb9275 : SUCCESS in 44m 02s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/69/305369/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9578a74 : SUCCESS in 1h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ff41b4fd8bb0cf4c4eb5ae6d1155f171b9080a49","author":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"date":"2016-09-08 09:55:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"15ae49eadc1d3e1f48ca4771a21bf193c98b0d64","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-08 10:17:18.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/305369/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"d04eedbdaf969544b5aa7564000bacbb2ee96a2f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-08 10:56:13.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/69/305369/7/check/check-ibm-tempest-dsvm-full/eb345d1/ : SUCCESS in 1h 00m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe757e908cc4461c995276f6bf3313241a5ebee4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-08 11:10:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/69/305369/7/check/gate-nova-docs-ubuntu-xenial/a396d8f//doc/build/html/ : SUCCESS in 3m 31s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-nova-pep8-ubuntu-xenial/d1e7940/ : SUCCESS in 6m 28s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-nova-python27-db-ubuntu-xenial/a10cf9a/ : SUCCESS in 12m 05s\n- gate-nova-python34-db http://logs.openstack.org/69/305369/7/check/gate-nova-python34-db/ee0001c/ : SUCCESS in 15m 03s\n- gate-nova-python35-db-nv http://logs.openstack.org/69/305369/7/check/gate-nova-python35-db-nv/ef9c1b3/ : SUCCESS in 13m 10s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/69/305369/7/check/gate-grenade-dsvm-neutron-ubuntu-trusty/46799ac/ : SUCCESS in 45m 49s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/edd5dbc/ : SUCCESS in 1h 07m 24s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-full-ubuntu-xenial/e28bbc8/ : SUCCESS in 50m 25s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/69/305369/7/check/gate-grenade-dsvm-ubuntu-trusty/f0fe576/ : SUCCESS in 37m 22s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-nova-tox-db-functional-ubuntu-xenial/1fc1a5a/ : SUCCESS in 14m 09s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/69/305369/7/check/gate-grenade-dsvm-multinode/f1895c1/ : SUCCESS in 45m 04s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/84feac8/ : SUCCESS in 22m 50s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-cells/1592ed6/ : SUCCESS in 35m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph/25ed261/ : SUCCESS in 52m 36s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/3e18abb/ : SUCCESS in 1h 12m 41s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/43681b7/ : SUCCESS in 1h 02m 56s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/69/305369/7/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/c776d55/ : SUCCESS in 27m 16s","accounts_in_message":[],"_revision_number":7},{"id":"6d54a4228f45482639e605e46f0832344639b55c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 11:10:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/69/305369/7/check/dsvm-tempest-kvm/1d2faa6 : SUCCESS in 1h 14m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"1146f6c5e7a150fd75017ff301fdbde2c25010d2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 11:49:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/69/305369/7/check/check-dsvm-tempest-vz7-exe-minimal/02cab42 : SUCCESS in 1h 53m 08s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"4e6273994469169ae6e8dcf7cddb979de8c19f32","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 12:00:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/305369/7/ : SUCCESS in 2h 05m 21s","accounts_in_message":[],"_revision_number":7},{"id":"1e746b8af10e8d7e738ec5c45cebf3b7a2c333d3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-08 12:12:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/7/check/dsvm-tempest-nova-network/9afceb6 : SUCCESS in 1h 23m 00s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/69/305369/7/check/dsvm-tempest-neutron-network/0da0689 : SUCCESS in 1h 41m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"977f8c3883423a826032e6557b8ee774d0a099ed","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 12:40:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/69/305369/7/check/tempest-dsvm-full-nfv-xenial/677336f : SUCCESS in 1h 54m 23s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/69/305369/7/check/tempest-dsvm-intel-nfv-xenial/84f93a0 : SUCCESS in 47m 04s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/69/305369/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4301395 : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":7},{"id":"74a960bdca1d0db4c36250f8d127cf2546088901","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-08 18:09:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/69/305369/7/check-nova/Nova-ML2-Sriov/f82af7e : SUCCESS in 40m 42s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/69/305369/7/check-nova/Nova-MACVTAP-ML2-Sriov/c25dfb7 : SUCCESS in 34m 21s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"11c5b091e94347021c70001954f4e449858453c0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-09 08:21:35.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/305369/7/808/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"929ccedfe1751d5e28eb042dec73e5bb0bdc863d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-21 02:52:27.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"80b6e3e8339de651b42f2ea9fb473ea9753e7240","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-10-04 23:43:31.000000000","message":"Patch Set 7: Code-Review-1\n\nThis direction is good for me, and I prefer to help that.\n\nbut I feel the spec is necessary for this to investigate current problems of non-accurate success codes before the code implementation.\n\nIt would be nice to avoid random changes for accurate success codes with a lot of microversions because Ocata is short dev cycle and conflicts with the other patches (which require microversion bumps) would be critical.","accounts_in_message":[],"_revision_number":7},{"id":"c8e9c6ef1f7a0b1c78f39cfc6f695b2ad6b7d1be","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-11-23 22:53:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c72fab1af27702141d921b9694a03bf6330ee9b3","revisions":{"2cbee8abaf3125dfa2204a3df46509c6d69cc6ce":{"kind":"REWORK","_number":1,"created":"2016-04-13 15:57:37.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/1"}}},"commit":{"parents":[{"commit":"b9d757bc0429159a235a397c51d510bd40e19709","subject":"Merge \"Remove unused parameter from _get_requested_instance_group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b9d757bc0429159a235a397c51d510bd40e19709"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"subject":"Add versioning to response","message":"Add versioning to response\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2cbee8abaf3125dfa2204a3df46509c6d69cc6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2cbee8abaf3125dfa2204a3df46509c6d69cc6ce"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b9d757bc0429159a235a397c51d510bd40e19709","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"87083c98e530cfc521b38b561a23918a401b33bd":{"kind":"REWORK","_number":2,"created":"2016-04-19 14:33:07.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/2"}}},"commit":{"parents":[{"commit":"b9d757bc0429159a235a397c51d510bd40e19709","subject":"Merge \"Remove unused parameter from _get_requested_instance_group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b9d757bc0429159a235a397c51d510bd40e19709"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-19 14:33:27.000000000","tz":180},"subject":"Add versioning to response","message":"Add versioning to response\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87083c98e530cfc521b38b561a23918a401b33bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87083c98e530cfc521b38b561a23918a401b33bd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b9d757bc0429159a235a397c51d510bd40e19709","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5718dd6a9bea5c4a42a8ed22fc738c89cb02ca6f":{"kind":"REWORK","_number":3,"created":"2016-06-07 16:20:34.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/3"}}},"commit":{"parents":[{"commit":"e5baa9df9a2959828cecf9546209522bbc1f348b","subject":"Merge \"Fix nova-compute start failed when reserved_huge_pages has value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e5baa9df9a2959828cecf9546209522bbc1f348b"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-06-07 16:18:45.000000000","tz":180},"subject":"Add versioning to response","message":"Add versioning to response\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5718dd6a9bea5c4a42a8ed22fc738c89cb02ca6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5718dd6a9bea5c4a42a8ed22fc738c89cb02ca6f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e5baa9df9a2959828cecf9546209522bbc1f348b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3cf7d80c1adccbaf91930cd14ebc1c9a2f4d8638":{"kind":"REWORK","_number":4,"created":"2016-06-24 11:25:40.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/4"}}},"commit":{"parents":[{"commit":"6a119fc72096fa65678112e151e4a7a00091ce26","subject":"Merge \"XenAPI: OVS agent updates the wrong port with Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a119fc72096fa65678112e151e4a7a00091ce26"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-06-24 11:24:15.000000000","tz":180},"subject":"Pin HTTP response codes to API microversions","message":"Pin HTTP response codes to API microversions\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cf7d80c1adccbaf91930cd14ebc1c9a2f4d8638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cf7d80c1adccbaf91930cd14ebc1c9a2f4d8638"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6a119fc72096fa65678112e151e4a7a00091ce26","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3e331ecd6914bd02de1eb9f77235661d34a8484e":{"kind":"REWORK","_number":5,"created":"2016-06-27 13:45:10.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/5"}}},"commit":{"parents":[{"commit":"6a119fc72096fa65678112e151e4a7a00091ce26","subject":"Merge \"XenAPI: OVS agent updates the wrong port with Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a119fc72096fa65678112e151e4a7a00091ce26"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-06-27 13:44:44.000000000","tz":180},"subject":"Pin HTTP response codes to API microversions","message":"Pin HTTP response codes to API microversions\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e331ecd6914bd02de1eb9f77235661d34a8484e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e331ecd6914bd02de1eb9f77235661d34a8484e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6a119fc72096fa65678112e151e4a7a00091ce26","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bc05d36a26177eb2d760312e5e5aa6b1435aad49":{"kind":"REWORK","_number":6,"created":"2016-06-29 12:57:04.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/6"}}},"commit":{"parents":[{"commit":"6a119fc72096fa65678112e151e4a7a00091ce26","subject":"Merge \"XenAPI: OVS agent updates the wrong port with Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a119fc72096fa65678112e151e4a7a00091ce26"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-06-29 12:56:24.000000000","tz":180},"subject":"Pin HTTP response codes to API microversions","message":"Pin HTTP response codes to API microversions\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc05d36a26177eb2d760312e5e5aa6b1435aad49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc05d36a26177eb2d760312e5e5aa6b1435aad49"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6a119fc72096fa65678112e151e4a7a00091ce26","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c72fab1af27702141d921b9694a03bf6330ee9b3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2016-09-08 09:55:04.000000000","uploader":{"_account_id":12712,"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","username":"stgleb"},"ref":"refs/changes/69/305369/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/305369/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/305369/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/305369/7"}}},"commit":{"parents":[{"commit":"b028b253913b0129b205dd279242c66936652998","subject":"Merge \"Refresh info_cache after deleting floating IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b028b253913b0129b205dd279242c66936652998"}]}],"author":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-04-13 15:57:44.000000000","tz":180},"committer":{"name":"Gleb Stepanov","email":"gstepanov@mirantis.com","date":"2016-09-08 09:55:04.000000000","tz":0},"subject":"Pin HTTP response codes to API microversions","message":"Pin HTTP response codes to API microversions\n\nSome methods could have different response codes depending\non version, like EvacuateController._evacuate should return\n202 code instead of 200 because it is asynchronous, this new\ndecorator enables to change response codes in new versions\npreserving backward compatibility.\n\nChange-Id: I2c081878c18f2905a9509cfdbc84deec32127154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c72fab1af27702141d921b9694a03bf6330ee9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c72fab1af27702141d921b9694a03bf6330ee9b3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b028b253913b0129b205dd279242c66936652998","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
