)]}'
{"id":"openstack%2Fnova~312051","triplet_id":"openstack%2Fnova~master~Iba73e582d36fa16f563a8e806073a06f857e1078","project":"openstack/nova","branch":"master","topic":"bug/1577758","hashtags":[],"change_id":"Iba73e582d36fa16f563a8e806073a06f857e1078","subject":"fix wrong key name in test code","status":"MERGED","created":"2016-05-03 13:07:33.000000000","updated":"2016-05-05 01:55:34.000000000","submitted":"2016-05-05 01:55:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d685919ab8a0f7a0d7e3d925de437a5be4905be","_number":312051,"virtual_id_number":312051,"owner":{"_account_id":21646,"name":"Peng Li","email":"flyd1005@outlook.com","username":"flyd1005"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2016-05-04 15:38:40.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2016-05-04 11:02:01.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-05-05 01:55:33.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-05-04 10:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-05-04 16:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":1,"date":"2016-05-04 18:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},{"value":2,"date":"2016-05-05 01:05:57.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},{"value":1,"date":"2016-05-05 01:05:57.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-05-03 13:07:45.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2016-05-03 14:17:34.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-03 16:10:23.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-03 18:22:55.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-04 10:02:28.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-04 11:02:01.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-04 15:38:40.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-05-04 16:20:01.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2016-05-04 18:38:38.000000000","updated_by":{"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},"reviewer":{"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},"state":"REVIEWER"},{"updated":"2016-05-05 01:05:57.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2016-05-05 01:55:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9a67354d3dc822220a0a92e75248431c75db8d4e","author":{"_account_id":21646,"name":"Peng Li","email":"flyd1005@outlook.com","username":"flyd1005"},"date":"2016-05-03 13:07:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70b6d6a353c637430d9f3ab4f395e61f9c464608","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2016-05-03 13:07:45.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. Also it\n    takes generally at least a couple of weeks for cores to get around to\n    reviewing code. The more you participate in the community the more\n    rewarding it is for you. You may also notice that the more you get to know\n    people and get to be known, the faster your patches will be reviewed and\n    eventually approved. Get to know others and become known by doing code\n    reviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: http://docs.openstack.org/infra/manual/developers.html\n    Commit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"b47b8be372158383cbfeaceee7fd3382220f7e50","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-03 13:24:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/51/312051/1/check/gate-nova-docs/a937774//doc/build/html/ : SUCCESS in 3m 23s\n- gate-nova-pep8 http://logs.openstack.org/51/312051/1/check/gate-nova-pep8/3960845/ : SUCCESS in 6m 04s\n- gate-nova-python27-db http://logs.openstack.org/51/312051/1/check/gate-nova-python27-db/ea98532/ : SUCCESS in 8m 14s\n- gate-nova-python34-db http://logs.openstack.org/51/312051/1/check/gate-nova-python34-db/5db23df/ : SUCCESS in 10m 00s\n- gate-nova-tox-db-functional http://logs.openstack.org/51/312051/1/check/gate-nova-tox-db-functional/4c5fb57/ : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c43b6e10115a2027b17e28afa2de364e3970b61f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-03 13:24:59.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/312051/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/ff53f70380f146509a22cf5ef463b23f/index.html : SUCCESS in 9m 54s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/312051/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/b412463c3ec8448892c44f068d8ab754/index.html : SUCCESS in 12m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/312051/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/595710e5c7904beba9318648ecd368a3/index.html : SUCCESS in 11m 27s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/312051/1/check/real-db-upgrade_nova_percona_user_001:th-percona/0b492ab03ced47d78b2e1c9e45f09d1b/index.html : SUCCESS in 9m 19s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e0ea906dfafe34e9f954ebe7aae631be8d1b2ac","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-03 14:17:34.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/312051/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d3a1c0e5e566442e69dfc61a8287b7d38475aae1","author":{"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},"date":"2016-05-03 15:53:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease update commit message. Otherwise lgtm.","accounts_in_message":[],"_revision_number":1},{"id":"b664b19defbcbb58a1c5fc3e73955697f4488db3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-03 16:10:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/312051/1/ : FAILURE in 3h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"84b666d09754e5a3ab0f1251b677cb11119ae092","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-03 18:22:55.000000000","message":"Patch Set 1:\n\nBuild Successful.\n* check-tempest-compute http://intel-openstack-ci-logs.ovh/compute-ci/refs/changes/51/312051/1 : SUCCESS\n\nTo re-run the job please post a comment with `intel-nfv-ci recheck`","accounts_in_message":[],"_revision_number":1},{"id":"2ffb512f0a5c5f0c3f8e22fc1bc78d752ec49800","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-03 18:25:40.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/51/312051/1/13692/index.html : SUCCESS in 01 h 07 m 54 s","accounts_in_message":[],"_revision_number":1},{"id":"2ef1502b834443b5ee2d44161891b73b5410612a","author":{"_account_id":21646,"name":"Peng Li","email":"flyd1005@outlook.com","username":"flyd1005"},"date":"2016-05-04 09:47:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"538a94f011c14bbdae584535c8df9a1be9beffe1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-04 09:47:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"ccbe387e628f2c1c20d6ad94cd4454499ebe3cfe","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-04 10:02:28.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/51/312051/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e705981f4fbd4430aa09db98a69f1a17/index.html : SUCCESS in 13m 46s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/51/312051/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/65827d3abd6d42e6b3fa083f97c4ebb7/index.html : SUCCESS in 10m 35s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/51/312051/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/fcb18cd36c204720b456b8bde94f3c92/index.html : SUCCESS in 10m 33s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/51/312051/2/check/real-db-upgrade_nova_percona_user_001:th-percona/ee4f7998d5ff4445ab99e6c84bff38d2/index.html : SUCCESS in 9m 15s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"5dcf82da0bf29efab11861a288c590dc81c07676","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-04 10:02:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/51/312051/2/check/gate-nova-docs/3b56262//doc/build/html/ : SUCCESS in 3m 15s\n- gate-nova-pep8 http://logs.openstack.org/51/312051/2/check/gate-nova-pep8/1899114/ : SUCCESS in 5m 14s\n- gate-nova-python27-db http://logs.openstack.org/51/312051/2/check/gate-nova-python27-db/905308a/ : SUCCESS in 7m 52s\n- gate-nova-python34-db http://logs.openstack.org/51/312051/2/check/gate-nova-python34-db/677e4d3/ : SUCCESS in 10m 54s\n- gate-nova-tox-db-functional http://logs.openstack.org/51/312051/2/check/gate-nova-tox-db-functional/fca4775/ : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b937af0fc62de8762dd17f977a2df9b6e7d60249","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-04 10:20:18.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/312051/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ce8666780c887e254233e4005c70ee2cffd52427","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-04 11:02:01.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/51/312051/2/13798/index.html : SUCCESS in 01 h 12 m 13 s","accounts_in_message":[],"_revision_number":2},{"id":"a623576c98ae98269f4e256d6d960bafc69778a0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-04 12:04:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/312051/2/ : SUCCESS in 2h 17m 17s","accounts_in_message":[],"_revision_number":2},{"id":"680ca37465d9fe1425d031e9129dd33094cb3c0a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-05-04 15:38:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/312051/2/4415 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fdcc7117a3434559c6b41b99146b283453d53f08","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-05-04 16:20:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d64942b59c692740f3f86c8b2566a757be9e9493","author":{"_account_id":18335,"name":"Pushkar Umaranikar","email":"pushkar.umaranikar@intel.com","username":"pushkarumaranikar"},"date":"2016-05-04 18:38:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4ac1faa2c6b342619842ab7506ccbd5992dbd636","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2016-05-05 01:05:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"df297a493dc203767a3e3d92c4950e6c1778b45c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-05 01:06:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"4c6f1340f8e3c386def1327424def8b7dd78156f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-05 01:55:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/51/312051/2/gate/gate-nova-docs/8dc3b99//doc/build/html/ : SUCCESS in 3m 38s\n- gate-nova-pep8 http://logs.openstack.org/51/312051/2/gate/gate-nova-pep8/6a77bd5/ : SUCCESS in 5m 55s\n- gate-nova-python27-db http://logs.openstack.org/51/312051/2/gate/gate-nova-python27-db/28f8cb5/ : SUCCESS in 7m 38s\n- gate-nova-python34-db http://logs.openstack.org/51/312051/2/gate/gate-nova-python34-db/7f00501/ : SUCCESS in 7m 51s\n- gate-nova-tox-db-functional http://logs.openstack.org/51/312051/2/gate/gate-nova-tox-db-functional/c9a41bb/ : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"958ace87aa058ffc4aa4ef53b12730f58e58b78d","date":"2016-05-05 01:55:34.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91bef29e1696d9c88ff1becff1f4d25abfbac5db","revisions":{"baa7e1403a056de1e19a2710e3efe62c35ee8a85":{"kind":"REWORK","_number":1,"created":"2016-05-03 13:07:33.000000000","uploader":{"_account_id":21646,"name":"Peng Li","email":"flyd1005@outlook.com","username":"flyd1005"},"ref":"refs/changes/51/312051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/312051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/312051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/312051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/312051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/312051/1"}}},"commit":{"parents":[{"commit":"04d13c8478f4c55d999a250f0ac555279450bf25","subject":"Merge \"libvirt: Fix ssh driver to to prevent prompting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04d13c8478f4c55d999a250f0ac555279450bf25"}]}],"author":{"name":"Peng Li","email":"flyd1005@outlook.com","date":"2016-05-03 12:43:35.000000000","tz":480},"committer":{"name":"Peng Li","email":"flyd1005@outlook.com","date":"2016-05-03 13:07:18.000000000","tz":480},"subject":"fix wrong key name in test code","message":"fix wrong key name in test code\n\ntest_default_root_device_name() in nova/tests/unit/virt/libvirt/test_driver.py,\n\u0027detination_type\u0027 should be \u0027destination_type\u0027.\n\nChange-Id: Iba73e582d36fa16f563a8e806073a06f857e1078\nCloses-Bug: #1577758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/baa7e1403a056de1e19a2710e3efe62c35ee8a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/baa7e1403a056de1e19a2710e3efe62c35ee8a85"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"04d13c8478f4c55d999a250f0ac555279450bf25","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"91bef29e1696d9c88ff1becff1f4d25abfbac5db":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2016-05-04 09:47:20.000000000","uploader":{"_account_id":21646,"name":"Peng Li","email":"flyd1005@outlook.com","username":"flyd1005"},"ref":"refs/changes/51/312051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/312051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/312051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/312051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/312051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/312051/2"}}},"commit":{"parents":[{"commit":"04d13c8478f4c55d999a250f0ac555279450bf25","subject":"Merge \"libvirt: Fix ssh driver to to prevent prompting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04d13c8478f4c55d999a250f0ac555279450bf25"}]}],"author":{"name":"Peng Li","email":"flyd1005@outlook.com","date":"2016-05-03 12:43:35.000000000","tz":480},"committer":{"name":"Peng Li","email":"flyd1005@outlook.com","date":"2016-05-04 09:44:21.000000000","tz":480},"subject":"fix wrong key name in test code","message":"fix wrong key name in test code\n\ntest_default_root_device_name() in\nnova/tests/unit/virt/libvirt/test_driver.py,\n\u0027detination_type\u0027 should be \u0027destination_type\u0027.\n\nChange-Id: Iba73e582d36fa16f563a8e806073a06f857e1078\nCloses-Bug: #1577758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91bef29e1696d9c88ff1becff1f4d25abfbac5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91bef29e1696d9c88ff1becff1f4d25abfbac5db"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"04d13c8478f4c55d999a250f0ac555279450bf25","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
