)]}'
{"id":"openstack%2Fnova~31561","triplet_id":"openstack%2Fnova~master~Ife5be9922db0742c8ee4f970517396be86597cce","project":"openstack/nova","branch":"master","topic":"bp/cells-cinder-support","hashtags":[],"change_id":"Ife5be9922db0742c8ee4f970517396be86597cce","subject":"Cells: Add support for global cinder","status":"MERGED","created":"2013-06-03 20:54:42.000000000","updated":"2013-06-26 20:50:46.000000000","submitted":"2013-06-26 20:50:44.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d4b30c1f7272264fc23c439d84aa6c9331bde49a","_number":31561,"virtual_id_number":31561,"owner":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":0,"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},{"value":2,"date":"2013-06-26 20:50:44.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":1,"date":"2013-06-11 22:11:24.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"all":[{"value":2,"date":"2013-06-24 16:13:26.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":1,"date":"2013-06-26 10:19:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":2,"date":"2013-06-26 19:47:28.000000000","_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-06-14 19:45:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-06-12 17:19:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":1,"date":"2013-06-18 03:07:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2013-06-24 18:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"all":[{"value":0,"date":"2013-06-24 16:13:26.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":1,"date":"2013-06-26 19:47:28.000000000","_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-06-12 17:19:54.000000000","_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"date":"2013-06-14 14:14:37.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"value":0,"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-11 22:11:24.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-06-14 14:14:37.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2013-06-14 19:45:42.000000000","updated_by":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"reviewer":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"state":"REVIEWER"},{"updated":"2013-06-18 03:07:50.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2013-06-24 16:13:26.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2013-06-24 18:06:25.000000000","updated_by":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"reviewer":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"state":"REVIEWER"},{"updated":"2013-06-26 10:19:49.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"state":"REVIEWER"},{"updated":"2013-06-26 19:47:28.000000000","updated_by":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"reviewer":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"state":"REVIEWER"},{"updated":"2013-06-26 20:50:44.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7596bb4181d1e385c50c9d734bd4e35594e21715","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-03 20:55:25.000000000","message":"Patch Set 1: Work In Progress\n\nI need to add some tests yet.","accounts_in_message":[],"_revision_number":1},{"id":"82e707cd3d50be65bceeb8b19844d648ca987e5a","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-03 21:08:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0da7342235cf48d38e8c7fe115f3b938cfe0c18e","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-03 21:09:06.000000000","message":"Patch Set 2: Work In Progress\n\nstill need some tests","accounts_in_message":[],"_revision_number":2},{"id":"faf472f91b4b69f53a601fa83910b5acaddd30e6","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-03 21:10:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9d68a0ddfbd7ba9f00c26936d70cbfa727e22eb","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-03 21:11:03.000000000","message":"Patch Set 3: Work In Progress\n\nneeds tests yet.","accounts_in_message":[],"_revision_number":3},{"id":"9a5af4f871420d6e06c3e0cb453055ccc753d523","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-03 21:59:02.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/31561/3/check/gate-nova-pep8/27631 : SUCCESS in 3m 14s\n- http://docs-draft.openstack.org/31561/3/check/gate-nova-docs/21851/doc/build/html/ : SUCCESS in 4m 21s\n- http://logs.openstack.org/31561/3/check/gate-nova-python26/23248 : SUCCESS in 13m 00s\n- http://logs.openstack.org/31561/3/check/gate-nova-python27/23341 : SUCCESS in 9m 02s\n- http://logs.openstack.org/31561/3/check/gate-tempest-devstack-vm-full/20188 : SUCCESS in 47m 27s\n- http://logs.openstack.org/31561/3/check/gate-tempest-devstack-vm-postgres-full/19895 : FAILURE in 48m 06s (non-voting)\n- http://logs.openstack.org/31561/3/check/gate-tempest-devstack-vm-quantum/27204 : SUCCESS in 20m 58s\n- http://logs.openstack.org/31561/3/check/gate-grenade-devstack-vm/7559 : SUCCESS in 14m 30s (non-voting)\n- http://logs.openstack.org/31561/3/check/gate-nova-pylint/12846 : SUCCESS in 9m 25s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"17b527ab7a5028f9f0b5017f809fbd0e3699a2d3","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-03 22:21:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"55959ba1700fdbcb5ff345e14998a2ac11832bdb","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-03 22:45:07.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/89103\n\tXenServer (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/89104\n","accounts_in_message":[],"_revision_number":4},{"id":"d75195ddc7685f7eb7af5630fc32f3c7945c3fb6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-03 23:13:46.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/31561/4/check/gate-nova-pep8/27637 : SUCCESS in 3m 18s\n- http://docs-draft.openstack.org/31561/4/check/gate-nova-docs/21857/doc/build/html/ : SUCCESS in 2m 32s\n- http://logs.openstack.org/31561/4/check/gate-nova-python26/23254 : SUCCESS in 13m 26s\n- http://logs.openstack.org/31561/4/check/gate-nova-python27/23347 : SUCCESS in 9m 37s\n- http://logs.openstack.org/31561/4/check/gate-tempest-devstack-vm-full/20207 : SUCCESS in 48m 07s\n- http://logs.openstack.org/31561/4/check/gate-tempest-devstack-vm-postgres-full/19914 : FAILURE in 52m 01s (non-voting)\n- http://logs.openstack.org/31561/4/check/gate-tempest-devstack-vm-quantum/27223 : SUCCESS in 27m 35s\n- http://logs.openstack.org/31561/4/check/gate-grenade-devstack-vm/7577 : SUCCESS in 14m 03s (non-voting)\n- http://logs.openstack.org/31561/4/check/gate-nova-pylint/12852 : SUCCESS in 9m 30s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"7dd30be38c776ee90ad65d55de7863ec7ac984f6","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2013-06-04 00:40:34.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\nAll works well for me","accounts_in_message":[],"_revision_number":4},{"id":"6ce9c12d82453812929c250562bffddc8a13d0e9","author":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"date":"2013-06-04 18:48:36.000000000","message":"Patch Set 4: (5 inline comments)\n\nMore questions than answers.  Enjoy. :)","accounts_in_message":[],"_revision_number":4},{"id":"fb54f29cbb89d59011d4772bfec138d139369728","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-04 18:57:25.000000000","message":"Patch Set 4: (5 inline comments)\n\n:)  Replies inline.","accounts_in_message":[],"_revision_number":4},{"id":"1ef119ddf2b422e6dd9e41edb373e5ddee27a9f1","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-04 20:02:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6950d0b0a613969d7b138e0acf5f44c3c0c66784","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-04 20:23:02.000000000","message":"Patch Set 5: Works for me\n\nSmokeStack Results (patch set 5):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/89215\n\tXenServer (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/89216\n","accounts_in_message":[],"_revision_number":5},{"id":"d06b426a85ca708d348a56ae74fa483e33dc4f80","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-04 20:56:07.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/31561/5/check/gate-nova-pep8/27741 : SUCCESS in 3m 22s\n- http://docs-draft.openstack.org/31561/5/check/gate-nova-docs/21961/doc/build/html/ : SUCCESS in 4m 26s\n- http://logs.openstack.org/31561/5/check/gate-nova-python26/23358 : SUCCESS in 21m 37s\n- http://logs.openstack.org/31561/5/check/gate-nova-python27/23451 : SUCCESS in 8m 54s\n- http://logs.openstack.org/31561/5/check/gate-tempest-devstack-vm-full/20460 : SUCCESS in 43m 34s\n- http://logs.openstack.org/31561/5/check/gate-tempest-devstack-vm-postgres-full/20167 : SUCCESS in 52m 31s (non-voting)\n- http://logs.openstack.org/31561/5/check/gate-tempest-devstack-vm-quantum/27489 : SUCCESS in 19m 10s\n- http://logs.openstack.org/31561/5/check/gate-grenade-devstack-vm/7821 : FAILURE in 17m 11s (non-voting)\n- http://logs.openstack.org/31561/5/check/gate-nova-pylint/12933 : SUCCESS in 9m 35s (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"130fc24e1190e6955bd3d479dc7c8b763c42be66","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2013-06-05 04:55:30.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"9a5ebd27694d0f70f1b37ce05a4bc77f911a1b36","author":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"date":"2013-06-05 15:58:27.000000000","message":"Patch Set 5: (8 inline comments)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"ef6d5847161b0b89c63c549185d72b4250712d0f","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-05 20:34:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f6cfdea841082d72c31cc4c66899acbcc0df3141","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-05 20:35:23.000000000","message":"Patch Set 6:\n\nFixed the RPC version conflict.\n\nBelliott: Probably should add healing at some point, \u0027just in case\u0027.  We have been running this for 6+ months or so without an issue so far... so I\u0027m inclined to punt that to a separate review.","accounts_in_message":[],"_revision_number":6},{"id":"a65e3e819e6062f6e703f22caa86b1b9124e5af9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-05 21:31:53.000000000","message":"Patch Set 6: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/31561/6/check/gate-nova-pep8/27939 : SUCCESS in 3m 22s\n- http://docs-draft.openstack.org/31561/6/check/gate-nova-docs/22158/doc/build/html/ : SUCCESS in 2m 46s\n- http://logs.openstack.org/31561/6/check/gate-nova-python26/23548 : FAILURE in 19m 03s\n- http://logs.openstack.org/31561/6/check/gate-nova-python27/23648 : FAILURE in 10m 32s\n- http://logs.openstack.org/31561/6/check/gate-tempest-devstack-vm-full/20855 : SUCCESS in 51m 33s\n- http://logs.openstack.org/31561/6/check/gate-tempest-devstack-vm-postgres-full/20562 : SUCCESS in 48m 14s (non-voting)\n- http://logs.openstack.org/31561/6/check/gate-tempest-devstack-vm-quantum/27929 : SUCCESS in 20m 31s\n- http://logs.openstack.org/31561/6/check/gate-grenade-devstack-vm/8200 : SUCCESS in 15m 04s (non-voting)\n- http://logs.openstack.org/31561/6/check/gate-nova-pylint/13056 : SUCCESS in 9m 47s (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"9cc1be69aa9c3a0bd4534c462126d11df994fbca","author":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"date":"2013-06-06 15:40:13.000000000","message":"Patch Set 6:\n\nTest fail looks legit. (1.8 !\u003d 1.9)","accounts_in_message":[],"_revision_number":6},{"id":"89460b298757a9cc71f8d10da180788623539a8d","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-06 15:50:43.000000000","message":"Patch Set 6:\n\nit\u0027s cool. i\u0027m an idiot and forgot to fix the tests.","accounts_in_message":[],"_revision_number":6},{"id":"52153e4479b5e28d12fad1a1651cbba9a760ee8d","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-06 15:51:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bf112cd331a90d5ec4d089a6353be0e46619ddfc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-06 15:51:37.000000000","message":"Patch Set 7: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"13b159e5eff36fb97e2074e60061162bb03ea73c","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-06 21:30:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"733c54e6c1fa8f7efca59ad196a2ec97243ca6cb","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-06 21:58:37.000000000","message":"Patch Set 8: Works for me\n\nSmokeStack Results (patch set 8):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/89648\n\tXenServer (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/89649\n","accounts_in_message":[],"_revision_number":8},{"id":"362b4ba50e6b03c981db2f03c40640df826bcc35","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-06 22:24:54.000000000","message":"Patch Set 8: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/31561/8/check/gate-nova-pep8/28098 : SUCCESS in 3m 22s\n- http://docs-draft.openstack.org/31561/8/check/gate-nova-docs/22317/doc/build/html/ : SUCCESS in 4m 27s\n- http://logs.openstack.org/31561/8/check/gate-nova-python26/23706 : SUCCESS in 20m 50s\n- http://logs.openstack.org/31561/8/check/gate-nova-python27/23806 : SUCCESS in 9m 17s\n- http://logs.openstack.org/31561/8/check/gate-tempest-devstack-vm-full/21263 : SUCCESS in 43m 44s\n- http://logs.openstack.org/31561/8/check/gate-tempest-devstack-vm-postgres-full/20970 : SUCCESS in 53m 24s (non-voting)\n- http://logs.openstack.org/31561/8/check/gate-tempest-devstack-vm-quantum/28364 : SUCCESS in 21m 53s\n- http://logs.openstack.org/31561/8/check/gate-grenade-devstack-vm/8557 : SUCCESS in 14m 12s (non-voting)\n- http://logs.openstack.org/31561/8/check/gate-nova-pylint/13186 : SUCCESS in 9m 38s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"71a10d50342d15834209903c9c448bb73b3c2c42","author":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"date":"2013-06-07 17:24:01.000000000","message":"Patch Set 8: (4 inline comments)\n\nA few things inline.","accounts_in_message":[],"_revision_number":8},{"id":"f0467333cf3a6d50ac805d5a9c7832fbaa4235b8","author":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"date":"2013-06-10 16:16:54.000000000","message":"Patch Set 8: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":8},{"id":"3ac36fe93845427a124329e822109eb75f5d50f5","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-11 04:24:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0af622fb50d731f16f113923734de9290d18e761","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-11 04:24:55.000000000","message":"Patch Set 9: Do not merge\n\nHad to rebase.  But I need to look this over a bit better now that some BDM changes have landed.  It may have broken my work here.","accounts_in_message":[],"_revision_number":9},{"id":"28f5e8add078a95a04dc61cb0b30ff979938f726","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-11 05:10:20.000000000","message":"Patch Set 9: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/31561/9/check/gate-nova-pep8/28392 : SUCCESS in 3m 32s\n- http://docs-draft.openstack.org/31561/9/check/gate-nova-docs/22611/doc/build/html/ : SUCCESS in 4m 45s\n- http://logs.openstack.org/31561/9/check/gate-nova-python26/23999 : SUCCESS in 24m 21s\n- http://logs.openstack.org/31561/9/check/gate-nova-python27/24100 : SUCCESS in 9m 58s\n- http://logs.openstack.org/31561/9/check/gate-tempest-devstack-vm-full/22128 : SUCCESS in 44m 54s\n- http://logs.openstack.org/31561/9/check/gate-tempest-devstack-vm-postgres-full/21835 : SUCCESS in 45m 28s\n- http://logs.openstack.org/31561/9/check/gate-tempest-devstack-vm-quantum/29354 : SUCCESS in 19m 11s\n- http://logs.openstack.org/31561/9/check/gate-grenade-devstack-vm/9300 : SUCCESS in 14m 58s (non-voting)\n- http://logs.openstack.org/31561/9/check/gate-nova-pylint/13418 : SUCCESS in 10m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"a8e627c6ced5a952aa6bcf9c81d9f2cf8ce93c89","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-11 12:00:06.000000000","message":"Patch Set 9: Works for me\n\nSmokeStack Results (patch set 9):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/90501\n","accounts_in_message":[],"_revision_number":9},{"id":"01354f9e7acc9d043d71ba1fa3ccacf5fb2425ec","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-11 21:11:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e738117ba1f7693595c098921cac2db54dfc39fa","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-11 21:30:37.000000000","message":"Patch Set 10: Works for me\n\nSmokeStack Results (patch set 10):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/90653\n","accounts_in_message":[],"_revision_number":10},{"id":"7da10671d58e6f9811d57eda37b742418b45ea08","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-11 21:50:47.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9e9283f112778b8fb769122a492e5c5b70e11395","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-11 22:11:24.000000000","message":"Patch Set 11: Works for me\n\nSmokeStack Results (patch set 11):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/90654\n","accounts_in_message":[],"_revision_number":11},{"id":"132ba09466fd1e70643a5b68076b19d061bd6540","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-11 22:51:05.000000000","message":"Patch Set 11: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/31561/11/check/gate-nova-pep8/28482 : SUCCESS in 5m 11s\n- http://docs-draft.openstack.org/31561/11/check/gate-nova-docs/22701/doc/build/html/ : SUCCESS in 6m 14s\n- http://logs.openstack.org/31561/11/check/gate-nova-python26/24090 : SUCCESS in 22m 22s\n- http://logs.openstack.org/31561/11/check/gate-nova-python27/24190 : SUCCESS in 11m 21s\n- http://logs.openstack.org/31561/11/check/gate-tempest-devstack-vm-full/22392 : SUCCESS in 51m 54s\n- http://logs.openstack.org/31561/11/check/gate-tempest-devstack-vm-postgres-full/22099 : SUCCESS in 58m 36s\n- http://logs.openstack.org/31561/11/check/gate-tempest-devstack-vm-quantum/29651 : SUCCESS in 23m 53s\n- http://logs.openstack.org/31561/11/check/gate-grenade-devstack-vm/9544 : SUCCESS in 20m 47s (non-voting)\n- http://logs.openstack.org/31561/11/check/gate-nova-pylint/13482 : SUCCESS in 15m 43s (non-voting)\n","accounts_in_message":[],"_revision_number":11},{"id":"4cde1502534e61c537a8fd8a33108b49b1ca287b","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-12 17:19:54.000000000","message":"Patch Set 11: No score\n\n","accounts_in_message":[],"_revision_number":11},{"id":"a78dc9abdaab537519a79d717839c9698ceb84d7","author":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"date":"2013-06-13 14:35:22.000000000","message":"Patch Set 11: I would prefer that you didn\u0027t merge this\n\nEarlier comments were ignored.","accounts_in_message":[],"_revision_number":11},{"id":"75b41e273916cfbbc3a88db52cbef5e89b428d4c","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-14 19:19:26.000000000","message":"Patch Set 11:\n\nShoot, not on purpose!  Looking now.","accounts_in_message":[],"_revision_number":11},{"id":"df1cad1301be69003c18db7afea24296bef0a7d8","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-14 19:26:25.000000000","message":"Patch Set 8: (4 inline comments)\n\nReplied in Patch Set 8.","accounts_in_message":[],"_revision_number":8},{"id":"93457c168b2fe28cd087dcbae006bbec6c9a05af","author":{"_account_id":5652,"name":"David Ripton","email":"dripton@redhat.com","username":"dripton"},"date":"2013-06-14 19:45:42.000000000","message":"Patch Set 11: Looks good to me, but someone else must approve\n\nSorry I missed inline replies earlier.","accounts_in_message":[],"_revision_number":11},{"id":"104758fc51110e3f6f61677877a1eb9d8532cbd6","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-18 01:31:47.000000000","message":"Patch Set 11:\n\nbump","accounts_in_message":[],"_revision_number":11},{"id":"2bfe1ff0025c04e2214791cb5327e4632df97dca","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2013-06-18 03:07:50.000000000","message":"Patch Set 11: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":11},{"id":"438a51e5bf7561a3e39c633a76a3bb2a74270c65","author":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"date":"2013-06-21 18:51:13.000000000","message":"Patch Set 11:\n\nbump","accounts_in_message":[],"_revision_number":11},{"id":"e5a2727bff1570cb2f4f22fbecd94efed316e0be","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2013-06-24 16:13:26.000000000","message":"Patch Set 11: Looks good to me (core reviewer)\n\n(1 inline comment)\n\nMinor inline comment about adding an additional log line.\n\nI see that healing was discussed in earlier patchsets.  I agree that healing would be nice, just in case, but am okay with leaving it out.  Ideally it wouldn\u0027t be necessary at all so maybe it\u0027s even better to leave it out and just address issues as they come up.","accounts_in_message":[],"_revision_number":11},{"id":"db5a2002464af0d67d58e3577966cdf2c9859447","author":{"_account_id":475,"name":"Rick Harris","email":"rick.harris@rackspace.com","username":"rconradharris"},"date":"2013-06-24 18:06:25.000000000","message":"Patch Set 11: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":11},{"id":"225f6a4d35762c2160862b653fea1bc322510cc4","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2013-06-26 10:19:49.000000000","message":"Patch Set 11: Looks good to me, but someone else must approve\n\ntested with two devstack nova cells, works well.","accounts_in_message":[],"_revision_number":11},{"id":"139124bde8ef5a61e949d77b729be7aa07322cf2","author":{"_account_id":2835,"name":"Brian Elliott","email":"bdelliott@gmail.com","username":"belliott"},"date":"2013-06-26 19:47:28.000000000","message":"Patch Set 11: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":11},{"id":"ed6fb827418d6378049fddea2f946b85a940cb37","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 19:48:51.000000000","message":"Patch Set 11: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":11},{"id":"2301e61377dcb55a4389e1de0d6f133f0969ae70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 20:50:44.000000000","message":"Patch Set 11: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/31561/11/gate/gate-nova-docs/24565/doc/build/html/ : SUCCESS in 5m 03s\n- http://logs.openstack.org/31561/11/gate/gate-nova-pep8/30347 : SUCCESS in 5m 07s\n- http://logs.openstack.org/31561/11/gate/gate-nova-python26/25950 : SUCCESS in 19m 38s\n- http://logs.openstack.org/31561/11/gate/gate-nova-python27/26053 : SUCCESS in 11m 04s\n- http://logs.openstack.org/31561/11/gate/gate-tempest-devstack-vm-full/26551 : SUCCESS in 1h 00m 43s\n- http://logs.openstack.org/31561/11/gate/gate-tempest-devstack-vm-postgres-full/26257 : SUCCESS in 52m 48s\n- http://logs.openstack.org/31561/11/gate/gate-tempest-devstack-vm-quantum/34209 : SUCCESS in 25m 06s\n- http://logs.openstack.org/31561/11/gate/gate-grenade-devstack-vm/13109 : FAILURE in 28m 26s (non-voting)\n","accounts_in_message":[],"_revision_number":11},{"id":"b26730e50403524959f1bf15806438294a831c2a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 20:50:46.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"39e8f5bd3bc976603c5a0a2dd4ada3dd6b79258d","revisions":{"db028d1885d25afb2b2a24cf18de7c4081a89d53":{"kind":"REWORK","_number":1,"created":"2013-06-03 20:54:42.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/1"}}},"commit":{"parents":[{"commit":"0afcc051a99443004510458da32e507385d58873","subject":"Merge \"In utils.tempdir, pass CONF.tempdir as an argument.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0afcc051a99443004510458da32e507385d58873"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:50:15.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/db028d1885d25afb2b2a24cf18de7c4081a89d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/db028d1885d25afb2b2a24cf18de7c4081a89d53"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0afcc051a99443004510458da32e507385d58873","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"987bc37ac936684b8c7de89918b492900606c453":{"kind":"REWORK","_number":2,"created":"2013-06-03 21:08:37.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/2"}}},"commit":{"parents":[{"commit":"0afcc051a99443004510458da32e507385d58873","subject":"Merge \"In utils.tempdir, pass CONF.tempdir as an argument.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0afcc051a99443004510458da32e507385d58873"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 21:05:40.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/987bc37ac936684b8c7de89918b492900606c453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/987bc37ac936684b8c7de89918b492900606c453"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0afcc051a99443004510458da32e507385d58873","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"48fbf4336f04c45b60703952c72f425987a7b37d":{"kind":"REWORK","_number":3,"created":"2013-06-03 21:10:22.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/3"}}},"commit":{"parents":[{"commit":"0afcc051a99443004510458da32e507385d58873","subject":"Merge \"In utils.tempdir, pass CONF.tempdir as an argument.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0afcc051a99443004510458da32e507385d58873"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 21:07:36.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48fbf4336f04c45b60703952c72f425987a7b37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48fbf4336f04c45b60703952c72f425987a7b37d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0afcc051a99443004510458da32e507385d58873","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1b8ac810070d31068a7cbe3c5db633aad88cf858":{"kind":"REWORK","_number":4,"created":"2013-06-03 22:21:16.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/4"}}},"commit":{"parents":[{"commit":"0afcc051a99443004510458da32e507385d58873","subject":"Merge \"In utils.tempdir, pass CONF.tempdir as an argument.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0afcc051a99443004510458da32e507385d58873"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 22:17:02.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b8ac810070d31068a7cbe3c5db633aad88cf858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b8ac810070d31068a7cbe3c5db633aad88cf858"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0afcc051a99443004510458da32e507385d58873","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e50369b27d0c31e53a5a175cf3ee0c3e8440e6af":{"kind":"REWORK","_number":5,"created":"2013-06-04 20:02:52.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/5"}}},"commit":{"parents":[{"commit":"0afcc051a99443004510458da32e507385d58873","subject":"Merge \"In utils.tempdir, pass CONF.tempdir as an argument.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0afcc051a99443004510458da32e507385d58873"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-04 19:59:48.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e50369b27d0c31e53a5a175cf3ee0c3e8440e6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e50369b27d0c31e53a5a175cf3ee0c3e8440e6af"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0afcc051a99443004510458da32e507385d58873","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a1650a5c36e2313abe401c7d1b6a02f9a27199f6":{"kind":"REWORK","_number":6,"created":"2013-06-05 20:34:04.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/6"}}},"commit":{"parents":[{"commit":"e9687d8bc7c5ba3ea4e348f61d2e658ad5099144","subject":"Merge \"Delete unused bin directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e9687d8bc7c5ba3ea4e348f61d2e658ad5099144"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-05 20:29:44.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a1650a5c36e2313abe401c7d1b6a02f9a27199f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a1650a5c36e2313abe401c7d1b6a02f9a27199f6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e9687d8bc7c5ba3ea4e348f61d2e658ad5099144","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a861bc1f93ca69d8082e710cb4ce0a31664254b0":{"kind":"REWORK","_number":7,"created":"2013-06-06 15:51:05.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/7"}}},"commit":{"parents":[{"commit":"e9687d8bc7c5ba3ea4e348f61d2e658ad5099144","subject":"Merge \"Delete unused bin directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e9687d8bc7c5ba3ea4e348f61d2e658ad5099144"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-06 15:48:16.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a861bc1f93ca69d8082e710cb4ce0a31664254b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a861bc1f93ca69d8082e710cb4ce0a31664254b0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e9687d8bc7c5ba3ea4e348f61d2e658ad5099144","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c0237c9fb1439ead034e923582e7bf39776ef6d3":{"kind":"REWORK","_number":8,"created":"2013-06-06 21:30:36.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/8"}}},"commit":{"parents":[{"commit":"6ed87355005f0797031cab7f57676860da5af820","subject":"Merge \"Add x-compute-request-id header when no response body\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ed87355005f0797031cab7f57676860da5af820"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-06 21:26:24.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0237c9fb1439ead034e923582e7bf39776ef6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0237c9fb1439ead034e923582e7bf39776ef6d3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6ed87355005f0797031cab7f57676860da5af820","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b7586cf363e053e3aff1359fd3feb11aa20a029e":{"kind":"REWORK","_number":9,"created":"2013-06-11 04:24:18.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/9"}}},"commit":{"parents":[{"commit":"a657af5fd50c8c3b4167c3bc99d9b2021062dc83","subject":"Merge \"Create an image BDM for every instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a657af5fd50c8c3b4167c3bc99d9b2021062dc83"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-11 04:09:07.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b7586cf363e053e3aff1359fd3feb11aa20a029e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b7586cf363e053e3aff1359fd3feb11aa20a029e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a657af5fd50c8c3b4167c3bc99d9b2021062dc83","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7bbd8fb08889ae17bdb89e10606574bf3eb22491":{"kind":"REWORK","_number":10,"created":"2013-06-11 21:11:43.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/10"}}},"commit":{"parents":[{"commit":"9b5e206027149861de86585690bc3d5141ec240e","subject":"Merge \"Rename functions in nova.compute.flavors from instance_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b5e206027149861de86585690bc3d5141ec240e"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-11 21:08:35.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bbd8fb08889ae17bdb89e10606574bf3eb22491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bbd8fb08889ae17bdb89e10606574bf3eb22491"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b5e206027149861de86585690bc3d5141ec240e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"39e8f5bd3bc976603c5a0a2dd4ada3dd6b79258d":{"kind":"REWORK","_number":11,"created":"2013-06-11 21:50:47.000000000","uploader":{"_account_id":1030,"name":"Chris Behrens","email":"cbehrens@codestud.com","username":"cbehrens"},"ref":"refs/changes/61/31561/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/61/31561/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/61/31561/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/61/31561/11"}}},"commit":{"parents":[{"commit":"9b5e206027149861de86585690bc3d5141ec240e","subject":"Merge \"Rename functions in nova.compute.flavors from instance_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b5e206027149861de86585690bc3d5141ec240e"}]}],"author":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-03 20:02:21.000000000","tz":0},"committer":{"name":"Chris Behrens","email":"cbehrens@codestud.com","date":"2013-06-11 21:47:42.000000000","tz":0},"subject":"Cells: Add support for global cinder","message":"Cells: Add support for global cinder\n\nImplements cinder support for compute cells when cinder is a global\ninstallation.  Adds syncing of the block device mapping table between\nchild cells and API cells.\n\nImplements blueprint cells-cinder-support\n\nChange-Id: Ife5be9922db0742c8ee4f970517396be86597cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39e8f5bd3bc976603c5a0a2dd4ada3dd6b79258d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39e8f5bd3bc976603c5a0a2dd4ada3dd6b79258d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b5e206027149861de86585690bc3d5141ec240e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
